)]}'
{"id":"openstack%2Fcinder~909513","triplet_id":"openstack%2Fcinder~master~Ib70219a9d085257c90a75ddcfcb935b4659fd28d","project":"openstack/cinder","branch":"master","topic":"new-location-api-support","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-07-28 19:05:20.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-07-30 01:36:45.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-07-29 14:07:00.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-07-30 01:36:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib70219a9d085257c90a75ddcfcb935b4659fd28d","subject":"Add support for glance new location APIs","status":"MERGED","created":"2024-02-20 09:07:55.000000000","updated":"2025-07-30 13:32:38.000000000","submitted":"2025-07-30 01:36:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"909513-new-location-api-support","meta_rev_id":"f5705689495b60df19d98d0e8d7d7a0b9a9813a6","_number":909513,"virtual_id_number":909513,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-30 01:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-07-29 14:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-07-29 14:07:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-07-29 14:14:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-20 09:08:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 09:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 10:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-20 10:23:54.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-20 10:24:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 11:32:20.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-20 11:34:23.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-20 11:47:04.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 11:47:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-20 12:06:05.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 12:57:40.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-20 14:20:13.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 17:55:09.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-06-24 16:08:51.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-06-24 18:02:34.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-08 04:19:23.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-15 18:02:46.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-15 18:18:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-15 20:48:47.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-16 12:52:03.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-12-18 22:48:50.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"CC"},{"updated":"2025-04-25 20:55:30.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-04-28 21:15:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-05-06 07:42:23.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"CC"},{"updated":"2025-05-06 11:09:27.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-17 23:05:09.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-05-17 23:21:54.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-05-17 23:38:36.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-05-17 23:39:21.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-07-28 13:23:10.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-07-29 14:15:08.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-07-30 13:31:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"578cce24f9ae79c2b8fdaed3ff561dfb07195be7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-20 09:07:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff4752113166d42ecfd2a3762582a17481714015","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 09:08:02.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16536/","accounts_in_message":[],"_revision_number":1},{"id":"9f2a7b0130634ca1805ac46aaa3286e51ecb914b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 09:08:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16536/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7ce391809a2febcdd4a496fabfad46ef1d147f04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 09:30:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b79f0876a244dec8a1e31243794544e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24ec13ac3b6146a6915c32ddb0fbb8b3 : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd5788ace74448edbaad7e765d3b77b5 : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a5a2e05052744a1bf3608da17fbb0d4 : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70bf043c309e3e5b864b3acc9bc14804de710e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 10:23:40.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/890509 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"94e15d342171eea69ca514fa76402eb39e4351bf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-20 10:23:54.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b9ec167b456c1e118161b3aa955265d4b0f35c27","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-20 10:23:59.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"c2a4dddd4c7e30a8952744ab52814059c1a541a5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-20 10:24:00.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"62b8c3d25ce8f4502c44d055aee76426669368d8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-20 11:32:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/752/.","accounts_in_message":[],"_revision_number":1},{"id":"081c6efdddcef81431d33d9cb408e90e88327d45","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-20 11:34:23.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-909513-1/FC/ : SUCCESS in 2h 26m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-909513-1/iSCSI/ : SUCCESS in 1h 53m 45s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"0fa9bb47322b1a7fb15a3b83ea103ab16683fe11","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-20 11:47:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/13/909513/1/ : SUCCESS in 1h 32m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e8d3041b8b405330a670ffc59898d376be1e637b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-20 11:47:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/13/909513/1/check/DellEMC_Unity_FC/3f2503a/DellEMC_Unity_FC/1367 : SUCCESS in 1h 18m 05s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/13/909513/1/check/DellEMC_Unity_iSCSI/e8a57a6/DellEMC_Unity_iSCSI/1397 : SUCCESS in 1h 26m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"67e0f46d084b581524127001d19ee51e31bc5d0d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-20 12:06:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/61f18292706c433d8f6a49dfdef520e7 : SUCCESS in 1h 20m 18s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0f1e7433ecfc42be9842705cc45b83c7 : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":1},{"id":"38003c976f9a4e77f331497f12f7c662958442c3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-20 12:57:40.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f4f238d28a33424f807f1e1458a28f55\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d2c3eb57083e4ced98045fd4a2263298 : FAILURE in 1h 28m 43s","accounts_in_message":[],"_revision_number":1},{"id":"788fec564199c52fe49b5e7dd237012bdcbe3d31","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-20 13:21:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/743/.","accounts_in_message":[],"_revision_number":1},{"id":"236ab993d4d7ab9b97de4061966bcbf1395f0ef4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-20 14:20:13.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/13/909513/1/check/DellEMC_PowerFlex_3.6/ccbeadc/DellEMC_PowerFlex_3.6/1498 : FAILURE in 3s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/13/909513/1/check/DellEMC_PowerFlex_4.5/d2b000e/DellEMC_PowerFlex_4.5/462 : FAILURE in 2h 28m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"58af0d4fe6fcce96c58b3cac146723dd151e57ce","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-22 17:55:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/13/909513/1/20240222232740/NEC-V-FC : SUCCESS in 3h 05m 58s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/13/909513/1/20240222232740/NEC-V-ISCSI : SUCCESS in 3h 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ebdf47209c9ca66909edb7dbc70f03e750705093","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-25 00:13:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0c249aa9ed1f48e28655b843f3ce1650 : SUCCESS in 1h 09m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9d02119f59e34327bdacfc37bb5da0b3 : FAILURE in 1h 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4426e05585a1f85355b68d123e4ea019095d3642","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 07:46:14.000000000","message":"Topic set to new-location-api-support","accounts_in_message":[],"_revision_number":1},{"id":"42a0a6b6c96e7fefccb87d182eeecc6dcbe97707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-24 15:22:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f21660a040ca99b9d2bedab45d39922cbb65fa58","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-24 15:25:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"be9de3adddb7a3ed13d32c046a4e2471e0aef477","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-24 15:58:50.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1331477bd2814f30a056f42065cd4b97\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e03f7bbfe46f4b5a89c48ec33a91a3bc : POST_FAILURE in 25m 27s","accounts_in_message":[],"_revision_number":2},{"id":"835341fb017464614e1e4a63e36ebb49fa50998a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-24 16:06:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909513 : FAILURE in 40m 27s","accounts_in_message":[],"_revision_number":2},{"id":"76eca16e59fe36e295af269e1734b0334dffe582","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-24 16:08:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/909513 : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c631d1ec7bce680bed08976a68b3b9feb486b8a4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-24 17:04:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/639/.","accounts_in_message":[],"_revision_number":2},{"id":"573a08f3a63d6f88c299b15c9f15d18c2d682191","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-24 17:09:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_Unity_FC/fbe4dcf/DellEMC_Unity_FC/2137 : SUCCESS in 1h 29m 20s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_Unity_iSCSI/8aa411d/DellEMC_Unity_iSCSI/2147 : SUCCESS in 1h 46m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"40ff9d86b12ff9f8d591c27ef0bdf9a536066645","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-24 17:21:00.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerFlex_3.6/8f91036/DellEMC_PowerFlex_3.6/1628 : SUCCESS in 1h 58m 04s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerFlex_4.5/1db68ae/DellEMC_PowerFlex_4.5/523 : SUCCESS in 1h 43m 10s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/3b5f796/DellEMC_PowerFlex_4.x_NVMeTCP/475 : FAILURE in 11m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"aebbc36825ae995cf3e495d4d3b8cd136814e919","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-24 18:02:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/13/909513/2/check/fujitsu-eternus-dx-iscsi/202d34e : SUCCESS in 2h 33m 48s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/13/909513/2/check/fujitsu-eternus-dx-fc/9b3cedd : FAILURE in 2h 39m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7bf312646b62d23a2e8e6529513931f8ece03d55","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-24 18:15:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-909513-2/FC/ : SUCCESS in 2h 52m 34s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-909513-2/iSCSI/ : SUCCESS in 1h 56m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"eb05d505212f9ef787daa0489cbc980524747ea0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-25 02:09:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/91282213ec0048c3ba97d7497322b372 : SUCCESS in 1h 19m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ac05f3f28afe4b328fdd494223313889 : SUCCESS in 1h 34m 44s","accounts_in_message":[],"_revision_number":2},{"id":"edad9cc30ce93473fa5529705dc2756f387f9156","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-08 04:19:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/13/909513/1/20240708044947/NEC-ISCSI : SUCCESS in 5h 12m 09s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/13/909513/1/20240708044947/NEC-FC : SUCCESS in 8h 29m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dbc97676d05adefce0cdbbb5faaf5ced31e3182e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-15 08:48:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ce6f5610976e671238949de7cac4ec757e41bd2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-15 09:15:38.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/54b9981533134a3aadb01d461df6b548\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/666cfac9858e48a99f2ade2ea1bb0840 : POST_FAILURE in 22m 32s","accounts_in_message":[],"_revision_number":2},{"id":"da6c16633b0dcd25b08a6417cc1ef49056ba52ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 09:26:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ffe0dfd0dfa49858f71be6a384fc8fd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f108a573b5194018a32becb26110d11a : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cf75c8d58ee54ede8c4380098c17bb44 : SUCCESS in 9m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a8811c429e3234a532946c4c6a0e3a0a66e59a5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-15 11:08:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7313e16ec0514401b34a01f3b56d3c64 : SUCCESS in 1h 08m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bc8ad87483994eee95f743898bbfceb7 : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":2},{"id":"1c86e537fdf678b684e6e622a8ee126e80a6d491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 17:56:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22091105dde04e82b44172150e05bfa2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9bcb3b9428443fb789abd367d28c55 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2a6026a10744ca092c3e0bff3ea612d : SUCCESS in 12m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32f613e283c9492d80809d9940d5cca5 : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85c20a00172443449b075d45ba35ce95 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6515055702e4803be6a057e21c6ccdd : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/600bf03ffc31478cac03bde781a7e34b : SUCCESS in 47m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02a406f865bc4415affdfe152a6562c8 : SUCCESS in 1h 26m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/897538d4850142f2a2a6b6f73af87d76 : SUCCESS in 58m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/45a67be3c4bf454ea5ed771283d8a6ab : SUCCESS in 17m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a842766e7e454df4add7719f8c68fcfb : SUCCESS in 6m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c9810a62f7b54ab9b39646a6c48c7d34 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67cae9b910494606b08486ae7de90d6b : SUCCESS in 7m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2e0996e7a7e14900ab36a9a11df21358 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/96c813b577d64526934206e6e49ee104 : FAILURE in 24m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be755094f6504d5898f96e6197f59aee : SUCCESS in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70ae43d519cf41aea2a977d26b76a73d : SUCCESS in 1h 41m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a2bfc5828af48089915dfa6b4e5bc4e : RETRY_LIMIT Host unreachable in 2h 20m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f81445410e994389abcf9821cb9d5e87 : SUCCESS in 1h 49m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/71a8da593f5a4d389e969d146e517406 : SUCCESS in 1h 53m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c7349994809e492f9cf0d22f1ac4c4ae : SUCCESS in 29m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6c4433f403d441e09d48b8f0019bf721 : SUCCESS in 53m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a2de387a50974f36b4f658723828caa6 : SUCCESS in 1h 42m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6d88453b8bd84e0d86ca975bb1c708e8 : SUCCESS in 39m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/98d6c73bc2664fa2ac2ccc97cc5b1615 : SUCCESS in 1h 13m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/270e6d1c1cce4b27a311f43c9c27a055 : SUCCESS in 1h 37m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19ecd3f4dcfc4f3e996f152a49daa0b9 : SUCCESS in 1h 32m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c45b17d404084d00984917cb24184aad : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0ab5379e88dda83aa16f2c7dc8424ce9a099eb77","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-15 18:02:46.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_07_2024_17_57_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"cc853944aa9aa5189de717130b7c6c990f7f6d85","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-15 18:18:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/909513/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/97697ed/ : RETRY_LIMIT in 9m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/909513/2/upstream-check/cinder-tempest-plugin-ontap-nfs/7ccfbc2/ : RETRY_LIMIT in 8m 49s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c5bceb1b87c234ff8f3ccc5c23b8f7afc6816e20","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-15 20:48:47.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerStore_iSCSI/246d6a9/DellEMC_PowerStore_iSCSI/1513 : SUCCESS in 1h 25m 37s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerStore_FC/8a8ab86/DellEMC_PowerStore_FC/1710 : SUCCESS in 1h 01m 15s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerStore_NFS/d2fde22/DellEMC_PowerStore_NFS/1208 : FAILURE in 1h 10m 35s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerStore_NVMeTCP/90d2f45/DellEMC_PowerStore_NVMeTCP/1060 : SUCCESS in 2h 01m 41s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerStore_NVMeFC/22374e3/DellEMC_PowerStore_NVMeFC/26 : FAILURE in 2h 51m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"1586ae825eda73e3bce6dc0345d71fdfabf2c471","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-15 21:56:15.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_Unity_FC/97174db/DellEMC_Unity_FC/2232 : SUCCESS in 1h 31m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_Unity_iSCSI/e27cee9/DellEMC_Unity_iSCSI/2241 : SUCCESS in 1h 51m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5f2031dfe4a63850b52db7946847a36048a99470","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-15 22:08:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerFlex_3.6/b58302d/DellEMC_PowerFlex_3.6/1717 : SUCCESS in 1h 56m 28s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerFlex_4.5/c369db8/DellEMC_PowerFlex_4.5/598 : SUCCESS in 2h 12m 53s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/13/909513/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/c9b4941/DellEMC_PowerFlex_4.x_NVMeTCP/553 : FAILURE in 13m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"ab983fee9fc94c5831995db4e2c7dbeba9bb9f96","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-16 12:52:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9e3107acbc33e748403ea2c1662681c5b7893727","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2024-12-18 22:48:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9bab661bf2144c78cce9e92fc7aef6b0776874c2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-25 20:55:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"da9977ea240c2e2a029cdd3b81f0de16bde138ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-28 19:18:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4e76116f3263cf761196e004a1788fd997bc07eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-28 19:25:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e51890a29cfaa2e902908c4fd30114683cfbd02c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-28 19:27:49.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"480649518bb67ef043a604fdfe738afac7f8a51d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:08:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d65fff32007c4213bba47d061512316c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb266913e30f4658b54c7b39cb5a7acf : SUCCESS in 17m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/215a044c8fd242fea3061d37db6b2e98 : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/364e9c55b91b4484a7f909d70925283c : SUCCESS in 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"53ef278a9cfad62eb63839d65af6e16197db97ab","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-28 21:06:46.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/667/.","accounts_in_message":[],"_revision_number":4},{"id":"43ab6f082f13c2168d53efb2bb194e0bf0a9f38c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-28 21:15:50.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/42369f42aff246548500700a568cb366\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6cb6b5f85b934db3a076a03814807daa : FAILURE in 55m 42s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5a3937c96e84418d869b493034d8ffa0 : FAILURE in 1h 19m 58s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc347faacb2f401db2c57802fe09b71a : FAILURE in 1h 49m 17s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b60f57691af74a0798cf800b6bd82342 : SUCCESS in 38m 40s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/66d2c5f401a3471fa183b75334b8dd4d : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b257227e95e04798a5ab2747223a0536 : NODE_FAILURE Node request 200-0000046393 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46f3ebcd8a2144c08548550c8ae5527c : RETRY_LIMIT in 4m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0af027c2eaac42519025c56fa76d29a4 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"e567eb9f724cdb69201b8392b33624f5286c9a50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-28 21:26:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909513 : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":4},{"id":"f2010ed3f2ba8e42ebd804bbf76f7a972091aee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 21:42:51.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5871d13c3084a4da3c81ff933c2e3e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dddc47b8a1894bdfb06fac5d7a389b1c : FAILURE in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9959835f0f164205b04639e80dc1bec6 : SUCCESS in 9m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44018d6e74404fb798381674f407a3a7 : SUCCESS in 10m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43b74e7f69444133bf34db160ce0e624 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90f64d76f335434ca250a35ee7a33896 : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/1ffeea52687d4850bf72c438b856557c : SUCCESS in 50m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7c349e27a59a4a6c916d0a2885050c7b : SUCCESS in 52m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a38cc244f857420381ff6a22e93932a4 : SUCCESS in 1h 43m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9ecd20dba8e4d62a7219ada86871053 : SUCCESS in 1h 00m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/269e71ba5a894d8fb97b1f0fa19d0561 : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fc338a4441b94e25aa7c4926869edc3b : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3b22ff51a704c299c88c3e29cc62565 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0d9c32c92d444f0db7a207171c6ecab5 : FAILURE in 3m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/43021790d5f84f83b1479ce9b12116f1 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4201405711a44834a71fc6b40ef51835 : FAILURE in 36m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/822d092ff95b4c4cbabc0b83c4b97d3a : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4412081e358c4625a159daf9f30cd9c2 : FAILURE in 2h 11m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fc8c077f95fc42fb89373a7d636aad2d : FAILURE in 2h 09m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/43bc3628f9334f7c92e949abfe470b1e : FAILURE in 1h 43m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/07b430c8c3c14e3f8e950b3ba927e66d : FAILURE in 17m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b4a5770967414cafb2c9de1d29708360 : SUCCESS in 16m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/92e27ad31fc2468bb681cd7f4db5ec7d : SUCCESS in 1h 06m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9ee02e05a6e345e9b9d9f1f4f6c728ef : SUCCESS in 1h 10m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5c08e888dffd4a6f97a7349973293710 : SUCCESS in 25m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4ebaa8b7c7cc4d1d888994dc77b30c0f : POST_FAILURE in 2h 11m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dad53f0aa2344b179091b72924794bd0 : FAILURE in 11m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/652f10b6502a447ba79d69ac6d7120c3 : SUCCESS in 1h 19m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d59cb72b09e2408cb1fc3d182f62b2d5 : SUCCESS in 1h 15m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09413b5875844d078ddb9aed9185b006 : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":4},{"id":"6056243b0b5744e0d44d0849562990f188c31048","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-28 23:03:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b97e2b115f4b4a26ac5ace38cd16e1a4 : SUCCESS in 1h 14m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3434c82d692543dc8ec444658f8288b8 : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":4},{"id":"683d46a315dadae23f283e60a59886c569549f7b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-29 00:03:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909513 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":4},{"id":"93eb285e782f86c35cb0bfb9ba0fa4501e16694a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-29 08:46:43.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/788/.","accounts_in_message":[],"_revision_number":4},{"id":"1a5f7c4013df88cb749734d7e938e05307d404d2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-29 21:44:11.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/13/909513/4/20250430051738/NEC-V-ISCSI : FAILURE in 51m 49s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/13/909513/4/20250430051738/NEC-V-FC : FAILURE in 1h 26m 30s","accounts_in_message":[],"_revision_number":4},{"id":"ffd7b4f0e70b94b17c238d598b9c41f1cf39b779","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-30 06:43:58.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/13/909513/4/20250430141344/NEC-ISCSI : FAILURE in 55m 01s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/13/909513/4/20250430141344/NEC-FC : FAILURE in 1h 30m 11s","accounts_in_message":[],"_revision_number":4},{"id":"55589ce4284009ae2543787b7406ec6d4ee2b801","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-06 08:04:56.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9baeddff100d4a60d24413ced1ccfd206e5bc19a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-06 09:56:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2052cb0dff8e77126a3453cee17e37a90dc21233","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-06 09:58:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"071ab57255174740ffd23b38e12b4a4db83072c9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-06 10:01:57.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/04f53bb8dfec41c581b5fb190172e40a : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/606a58376dc8443a948f2211c7015bc7 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"a0290a71386cecb1c48f554601dfeb22762c39f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-06 10:12:42.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/819/.","accounts_in_message":[],"_revision_number":5},{"id":"646ea9f6711596584ce53b63ebb120e80d614565","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-06 10:12:50.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/697/.","accounts_in_message":[],"_revision_number":5},{"id":"c8b8f567f1097045edcdd4138e20b639f235c526","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 10:19:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31c2bed8b1124bbebbbb1660804b97d5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0dff31d84bd4591b66fd03dbbf0c768 : SUCCESS in 21m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/acb14b98557447ce8af895957db07916 : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/71d8bb4a22a4451e9a2bba7667b11140 : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fb9ca9c4ac7fe725bde1e64ca62f69901faa7b36","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-06 10:48:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909513 : SUCCESS in 49m 41s","accounts_in_message":[],"_revision_number":5},{"id":"f28961cc60ad5e0a1bde6b294983f0c21d918f2e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-06 11:02:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909513 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":5},{"id":"2b3b611e6a3593fe01b68f65ff98f3bafd35d62a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-06 11:09:27.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"11346a7996c409288b807e64f299bbbd2ee94669","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-06 11:41:07.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e74cb3294c4943edbd7be61a7b5acdad\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b242834415374bd5a611b422a4de9901 : FAILURE in 53m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/883ecf1c716e44a9ae30120456222ffe : SUCCESS in 55m 35s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0e3db8ece924d739e0a811e186c60c5 : FAILURE in 1h 43m 55s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d8552e8268864588982fa5bd74804777 : SUCCESS in 43m 24s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7e9e40fe6434b0cbf098615cec453f7 : SUCCESS in 51m 34s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aaba12ef0d0549aa82921e30644b51bd : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e00b6b2f19d54f8aa253e9c9cd1eaeff : FAILURE in 1h 37m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b35e067fc853434a81eb28b24f22732a : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"b559023309016785d39f8ab5eae221d9289c3411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:51:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3d95bafc2b248429fe8cb8c7df86802\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3aca6e140654a779e0d783ec32d5c93 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa3d39016c304dcf964ade42dc0c5b4f : SUCCESS in 7m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e648346851924b80bdf1d7e713e0e97c : SUCCESS in 9m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d96211b701c481fb66a4c3fb8accb50 : SUCCESS in 13m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8592096848bd4743ba7e8ac5c8c50b34 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/f1880a78090747608d9bbbabfe5cabba : SUCCESS in 47m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9c91c3a1a234535828ab86257f484d3 : SUCCESS in 52m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/199427712e424187826f50501229e2be : SUCCESS in 1h 40m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37d473e9a4ce40a1ab8b5d4f795addb3 : SUCCESS in 58m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/69cfbff117a54aee87def5301646e274 : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/62981ccab37e4938a2816a67e6b340d4 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/882e2d4eac6d4dfdb0bcf26d60644413 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ca7694c2e534f1780c3148c2cea9225 : FAILURE in 3m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f982291aa90449a9a771bfbe5d534acc : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/efb5de57d6404a11b7ff2778ab3a71d4 : FAILURE in 32m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b9b358525214ee1b14a6c961052c340 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3e50bf822dc64a7b883b785ee95301e3 : FAILURE in 1h 58m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ee69b34fd81a47b3bb6712380f3ddab1 : FAILURE in 2h 42m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2153741c03d44a5aa0b898baf27dcc4d : FAILURE in 1h 50m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8837942bbe24475fa21c2a59a6549269 : FAILURE in 19m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2acc137ae87d432899b3cd92f7032cfa : SUCCESS in 30m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/72a75fc92dbc4037bbfdbf23c32ce2d7 : SUCCESS in 1h 06m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1588b2a505c54a3894b412db857f9210 : SUCCESS in 1h 34m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6e2d42e7cdbc4c72b9111771cb0c6eae : SUCCESS in 45m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b1b55e5b9a954ec78fb29c2e6ec923a2 : SUCCESS in 1h 22m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cd58b4e61b7d4bf6884e2464197aaf47 : FAILURE in 22m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a6036a7cf3a4c58b57f9c4b7d542c12 : SUCCESS in 1h 37m 43s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b522cc3b2cdc4050b4690685746d8d03 : SUCCESS in 1h 22m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70303b5a7aa14b4f95b7e9c31413b5e4 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":5},{"id":"0dd46caa8d2a9932927305926add30af7290d195","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-07 03:09:12.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/13/909513/5/20250507104325/NEC-V-ISCSI : FAILURE in 54m 17s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/13/909513/5/20250507104325/NEC-V-FC : FAILURE in 1h 25m 45s","accounts_in_message":[],"_revision_number":5},{"id":"d3545b44f139d43346f4881335a5fc0ded8bfd0e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-07 04:51:11.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/13/909513/5/20250507122028/NEC-ISCSI : FAILURE in 56m 15s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/13/909513/5/20250507122028/NEC-FC : FAILURE in 1h 30m 40s","accounts_in_message":[],"_revision_number":5},{"id":"75d286cc39b6d09cab6f8af37c85455bf0efb961","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-07 14:40:13.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b04fb08b7a6213396572365ed4204803b0ec5f23","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-13 06:02:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"764a77a8ae8fb18d9cd47a8624634b3f969c5554","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-13 06:08:02.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/687a1f69526345d3b4632b1bd229e364 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1d1fd1f0c1a34db2b4327a5a13472f64 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"b7f21d7de625020e04237e4a0ea95ad0bf75d4e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 06:21:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f425066813c4587954053dd9d7ac3f7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cdb2b58500f34bcab85a0b20a85cff2c : SUCCESS in 16m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/38f5797f750e4c289e57021574dce5c0 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/69442cd0770c4ccaa018dffe2c52f54f : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"742e3b63b21b6025803f1519bdaf097ca4bfa051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 08:26:57.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71e5a3f25d934e249027503def4c05ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b32bc3ee679412b904c7ce5f3c89a8b : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a81d29124501493bbd121e2fec347ffd : SUCCESS in 7m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/214d55d0e64b4aa7b6c1065442583b01 : SUCCESS in 9m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3769b8171cf841cfa03713f9d094849e : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d06064449a4704b886c8160202d18a : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/5dd429b9ccba4be8a8e065ddeefba308 : SUCCESS in 54m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6c284f677f249eb8a000a648cbf9fcf : SUCCESS in 40m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/826df84712b44e0ab32b4bbf92812dc8 : SUCCESS in 1h 23m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/821b2538f92e4293acaf916ed6621b9a : SUCCESS in 46m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea4987f833ab4bcea3f4df489caddc7e : SUCCESS in 21m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/71cc54c09cde427faa665ed4a412cc28 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7e0094944bb8460cb4db55ef8cea2b61 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cbbf7ecb255f4b51ac60e80e9ee384d9 : SUCCESS in 5m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/29931c23527847eab329c6bfe4671a7e : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2c7e76cd58134203bfe3585c3477bde8 : FAILURE in 26m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a230366aac58406aad2716af2a0f91ed : SUCCESS in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/44c36101aa314406a53bc6211bf0658a : FAILURE in 1h 57m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/00020a4e721544cb94f653bee88f998a : FAILURE in 2h 23m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7b524e3bebaa4884b04cbbcae11af4f9 : SUCCESS in 1h 45m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/549f0f7c4308404ab539694e699edd12 : FAILURE in 23m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/00c49e739c584a4989e6001930031d13 : SUCCESS in 34m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/13ca6092247e4511a4cc73e889a0a145 : SUCCESS in 57m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f52a9f1e18c9426cb8d3189f68314335 : SUCCESS in 1h 20m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ef08f72a40e40fc8a69b56ea9520545 : SUCCESS in 30m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2a7f40dd92fc4d5081d6558e4c3406be : SUCCESS in 1h 13m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3a995d8f55c448d6a15fa84edeea1b6b : FAILURE in 17m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f815edd8080443998effd8b5b79d5679 : SUCCESS in 1h 28m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaf0de68d2024982b7d632970882f91e : SUCCESS in 1h 06m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/892bc4771d494ef899377023a786187c : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":5},{"id":"10aff1dfe04f9600be5888a1f0edb5ab057c8465","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-16 07:24:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a0b31e8638d2f4bd3568ad196b2fb401e127e18e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 07:42:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/251bf2458f8c49a5b150aa1d09064c6c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9cb3a0931d849518b117103b50edcde : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/425f765a838546b99fb00a6527b5ca0b : SUCCESS in 15m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/32f74be983ff409ba2a520fa52a4f65c : SUCCESS in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7694aeb7fc8e86712e919173a440396b3ddc910d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-16 08:45:04.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ebc28b51f12a47fc8996efbd0bb41d69 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6545aff6940f401c831a7f47aab01408 : RETRY_LIMIT in 1h 14m 21s","accounts_in_message":[],"_revision_number":5},{"id":"f0bd3683795cb5af5f1e623db71553d5b91165ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:43:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d87d7a303afd4b4e984de059e101746d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89dc30b567c3404d8a87e670ee9c55cb : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b3843af91884eba9ad833c7759972aa : SUCCESS in 12m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4ef4946cdbe4a849d006640d3e4a148 : SUCCESS in 13m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51583519c69f481f8a9353da9ad3599b : SUCCESS in 16m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51667b2c6e8c4f49b3dffb5b90043e44 : SUCCESS in 16m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a0846263162c48a99da101a4e8d53f66 : SUCCESS in 55m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44fe2f8c78644f14927e87072338228e : SUCCESS in 48m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c54517769439454fbe7da581a1e0c5bd : SUCCESS in 1h 38m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02528e3a3da84c4197bb5cd8b85cb0ce : SUCCESS in 1h 06m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/072ebc62e8c94618bc7b94fcfb7e8cbe : SUCCESS in 25m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/09062cda98ad46289f5bf7ce10dbb0aa : SUCCESS in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc81c8cf70f44cef9fdbc8bcdd43f7ce : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/280a3b94c0b64f3593f92a1af53d2b85 : SUCCESS in 6m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/471d463bf4fb4f958bb21655a75282fa : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/76d386127bfa42f69fb0e28fb4b7b61e : FAILURE in 31m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5fc5ef8ccb1b491d98d524f53d1744da : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b08ac427b1524b5680bca0e0399b581e : SUCCESS in 1h 47m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ce031439cbc940b8855953fd04bf8771 : FAILURE in 2h 15m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b25b15fa63b481facf0a19a05b4b0d7 : SUCCESS in 1h 56m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/91f1e1e4fe5940cda4142aeefbf0db04 : FAILURE in 24m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4a5dc8a85ffd466c97168e5b6c96e58b : SUCCESS in 31m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a8a2d9a670b04c0db86927994c2cce58 : SUCCESS in 1h 09m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e8b89e80dfae49a18dcec7efa9a9406f : SUCCESS in 1h 25m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2f02c36302214b75913bfe325f91f1b9 : SUCCESS in 31m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/28aabfd3a25947548435f71f3572ca8e : SUCCESS in 1h 35m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8e4a55679ab44ad6b780422b67545904 : FAILURE in 23m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/125c3f1a820b4617ae999c0aa859d977 : SUCCESS in 48m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faa46b29f4874fdb8ffac0a7f9f56d97 : SUCCESS in 1h 34m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b40f23879a470e8f3e5d073b375af7 : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":5},{"id":"f6f68ecda550452244b881168c0b8754d62b0cc7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-05-16 11:51:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"aa1a89ab980d4c0c58a9edd051afa6799776f992","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-05-17 23:05:09.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"00c0a4ce804427857ec36e9bd1414307fc604e1e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-05-17 23:21:54.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"74b04817af9333f4efcd32493bd6f43a88f76c15","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-05-17 23:38:36.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"a038d176f60d677aa15f1dae651d8e5b5081efc0","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-05-17 23:39:21.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"93ef5aac45f982ca8722cb442649d11a3437e237","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-23 18:33:55.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"728129a16ff63e33fe5f4a0905222d9425cf9cd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 19:03:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/105d935d02b2462b8abd30bc5b4eee69\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54acbf7c4e934dd7a68ab8fd1e1636ef : SUCCESS in 21m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d8fee14226542e290b46736dafe6d59 : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/00b99dc961e647ac91fc6bf65a50aa9b : SUCCESS in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e8c866368aad5db35353b382c35e3433df1c15cb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-23 19:42:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2051346e2f424dfda94159020f92d7e4 : FAILURE in 59m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/acf9a8e33fbb4f09a8aef14a635ec7cd : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":6},{"id":"1ff6a7b21c5f2126a75033fb3515e929a4655724","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-23 19:45:21.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/493ec987953348b98ba3b27b9196734e\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aa5cf98dc6b24f78a1aabdd1b39966e0 : NODE_FAILURE Node request 200-0000076285 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e7e52b7b9a104263a3793da2674ff104 : NODE_FAILURE Node request 200-0000076286 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/55bc89882905490985784c573f0edc70 : SUCCESS in 57m 53s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d9ff04bf6f14bb79f30d5e15596073b : SUCCESS in 59m 57s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5466671b68c4951bbc62f1835faf0ac : FAILURE in 1h 08m 33s","accounts_in_message":[],"_revision_number":6},{"id":"991ab7dd1c34f2658860b075edcf506cf1c1b11f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-23 19:48:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1306/.","accounts_in_message":[],"_revision_number":6},{"id":"78bc4813c17f823e413081f22f13cf54eb88984c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-23 19:52:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909513 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":6},{"id":"67c56e43a5aaff4d23d8f8acf19d453ea86eb665","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-23 20:23:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1440/.","accounts_in_message":[],"_revision_number":6},{"id":"3c42f6deb98743135f11710c71ec48efbd70ea8d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-23 20:24:04.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-909513-6/FC/ : UNSTABLE in 1m 18s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-909513-6/iSCSI/ : FAILURE in 1h 49m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"b3ca95e7cfe21d221aba0256f2e924099f911bd7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-23 20:24:51.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"66509273543bf81ee67b5b182c81f6c08576d2d5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-23 20:44:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909513 : FAILURE in 1h 22m 55s","accounts_in_message":[],"_revision_number":6},{"id":"c8293fd1b923ed1807c537c6d53bc09c09a7949d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 21:01:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e3dd6a587e14dc2bb637c28772135bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352dfa6b0347420ebde30dd81c984ae6 : SUCCESS in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ab2ef973ba14db49493c327faef0d6e : SUCCESS in 12m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c76f6853c8a14f76bc2006806bcdcd7c : SUCCESS in 11m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb5df79be2db47719055c846ffd8127a : SUCCESS in 16m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d7126b2b8804e80baa422e9d879191d : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/0ada373a82b8436db2a0b28aa579a53a : SUCCESS in 1h 02m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e3929c22d989451ca77002dc60192ce2 : SUCCESS in 1h 04m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a5335805668498aa39321646757eaa0 : SUCCESS in 1h 22m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9500890d9407482f9f2921455892056c : SUCCESS in 1h 04m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c64a7f20353b4fbeaab71e2e7b8420a4 : SUCCESS in 26m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/55e06264296c448689be6fe4f06856de : SUCCESS in 6m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b97eec35c1474c86a676cdbe6340afbe : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/beaa54e0d2d541b884da5da4384dfd3a : SUCCESS in 7m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6f98fd7d5b9d492fa193beb456dcc971 : SUCCESS in 11m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9831118f3d864f82aee0501e78769d3f : FAILURE in 33m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/627ba5be11484e61a52f8e12457758ef : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/30d6f1583d6d4d2db13fc24cda1514c5 : SUCCESS in 1h 42m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ff95449c350243fd9033a7bea4dd01e5 : FAILURE in 2h 25m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94bebdf7a120421e9a23538c093f131e : SUCCESS in 1h 54m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/292c75c5ccf34f899b8e11a609ce6061 : FAILURE in 10m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/62ab7112a4fd41ffb4cbdab9140f2cd2 : SUCCESS in 16m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e7bd1e74c5b44b9ea9fdf833dc8565f4 : SUCCESS in 40m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e0e287a09e7468781956453682498d7 : SUCCESS in 1h 40m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ea58735300954a799bef917ab60cd60d : SUCCESS in 29m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b3d7e81b4676454daafd75f803f91d7d : SUCCESS in 50m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6f8f2b87fd5c48e586deac9d51756faa : FAILURE in 15m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7abd0875bf5346099a2689f830d1dba0 : SUCCESS in 1h 37m 29s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63ba9560d6854963af006e84ceb2a28f : SUCCESS in 1h 10m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d3afa78f21045bc87678b5bab11420c : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":6},{"id":"358bfe7b97746eaef24ae8552d14beefe9522369","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-23 21:07:22.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_07_2025_21_01_55\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"44b35b5969bc7eed31bca734452c51ef4899fbb3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:17:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"931ff4a8d035000780e906ede67b09da3630a4c9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-25 15:22:27.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6df391670432798e09fcd792e7b400fd1351fa0c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-26 00:47:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0d57ce4bd60d8aa04cc77f563dc9772fec2ec0d8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-28 13:23:10.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c568431a333f0c2a08e6c20dba6caca67b0ebbd5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-28 13:31:03.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"eac79d540513add7896722302a1102f136b99f9b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-28 13:31:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"214c1a06a525633c0375926395a299b6c7156b1d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-28 13:39:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"be3b8462048254539c6d59fa2901072c98e1e274","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-28 13:59:53.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"579a0e4bbdd9934ff4510a5620f241db1028b538","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-28 14:05:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"68df9d7585fe8105303d9f74e899662f0d397b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-28 14:26:43.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7c11b390a065548b57313d94718eab25a6e80507","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-28 14:27:42.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6b293d7df6c630d4672a98929348077edd41328e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:46:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170e6aefcda5495da2b7e28b3c982dd6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cae439b84c504f43a956b1e07fafb152 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4015313a7bb347708b1598b26cd0e0ba : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83b21c733f77435397b60e47f31ba259 : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"74f5f1901faca8e2652b9d06e52685b9496df5f9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-28 14:52:26.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1458/.","accounts_in_message":[],"_revision_number":7},{"id":"5b59b562f3b6047c72dd5c26e35ca04ccceb8e32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-28 15:18:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/73822a45471048c99323553541974911 : SUCCESS in 49m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/58c4370e20ac4e70a5a8943b576800b0 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":7},{"id":"00262f3bd224a361116eb32b79a2d5dd36e3308a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-28 15:22:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909513 : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":7},{"id":"d7df707cbf7df367564cc51c77b2a234a7c133aa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-28 15:27:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d456b3c51a754b429b6fff8e78dc6bd5\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/575c7c9b36ce497ab5fa10be770e3b0a : NODE_FAILURE Node request 200-0000076663 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/db1a9713757745ec9dd22755df98e53b : NODE_FAILURE Node request 200-0000076664 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/df7be6399a07488197e97d2d4d60b198 : SUCCESS in 59m 50s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/415edf46473c48b29f7b0bd12ff6080d : SUCCESS in 51m 16s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ba7c2687bb4413f96db51043919ab6a : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":7},{"id":"46be509c799ec66e9c951fb2270e89bd5957cb74","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-28 15:30:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1323/.","accounts_in_message":[],"_revision_number":7},{"id":"3a8e975bbd1a2de20c91cc36846a0e27900c93db","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-28 15:51:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909513 : FAILURE in 1h 23m 32s","accounts_in_message":[],"_revision_number":7},{"id":"3b3c190534f4854b929244a9d4d7662be21a54bc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-28 16:27:27.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"3a8ac160d1728b1539feedc7275ad3ccc259264e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-28 17:04:02.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-909513-7/FC/ : FAILURE in 2h 26m 33s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-909513-7/iSCSI/ : FAILURE in 2h 36m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"c5f644f6becd6492adfbe04f4b09cec1b13ec0d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 18:03:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa53eafe979a441cb86c2b9d8a6713a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35d1b054fa524d2c92e3d39582368f04 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c873d68d54774d89a356f37e64596577 : SUCCESS in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28a3fa5a00d84c62873349e2a142861d : SUCCESS in 12m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c42565c22b049b2a444efa3694faeb7 : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f51eb53c84b4c5c93691abe3b1716ed : SUCCESS in 15m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ab83611a7f7c4019877126ba7c9d81bd : SUCCESS in 1h 01m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/42d588653c264cd5aea30b88c3afa388 : FAILURE in 28m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ac11316fe45423aa5e5b1bfd0cb2b44 : SUCCESS in 1h 31m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af56504d209a4d02bb83de538a6bcb0e : SUCCESS in 1h 03m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b976d3afa5544ffc9795f94cca8d4a79 : SUCCESS in 24m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ce5c3612563f467c8e0dc8ad0d756a2d : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f80a3241ac4e49bc8b774ba394941690 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b66cddd94674ba28fdb43d68c5d1ee4 : SUCCESS in 3m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/366c17675b8a4986980374954f6179fc : SUCCESS in 8m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/74ff1a3f53264bd1bf6466992a411026 : FAILURE in 34m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b273c5988cfa4c42a3c29c25d6fdfa30 : SUCCESS in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/81ff3bd410744d3ab60afd37108850b9 : SUCCESS in 1h 51m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/26650feb629c47f3847ebd000aea0ba5 : FAILURE in 19m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/634be995312a43e78068c297a15419da : SUCCESS in 2h 12m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/31d9e8554e2b42a29ebf74e7a6e5ca35 : FAILURE in 18m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/935cf7f9faf041d3aa5286f3485a1335 : SUCCESS in 40m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1081834dd7a44e8ab12570b0a4bd321a : SUCCESS in 1h 10m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/62b822c2a9ff43bbac5ac1161a440be8 : SUCCESS in 1h 31m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75ab8d09cf464fd0b8d637eda81ccc3a : SUCCESS in 41m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/24c3ad4140f64716885aa5fe9ac43215 : TIMED_OUT in 3h 34m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eda138e3a2f04e42a015b5d591c30250 : FAILURE in 9m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51209df60f1640858bd4330f4e14dda4 : SUCCESS in 1h 06m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b114f39ae0ab4b93ba92dc014543f386 : SUCCESS in 1h 35m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc2ef76eef134735af8483478b83458c : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":7},{"id":"8e94231ad87feb3e14829ef41e366e545d47f266","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-28 19:05:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0d041890398c61ddd88f4301f87a58502057556f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 19:35:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff39b24811dd41c69ff0082ff1f119dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8fbc25dc4dd54173a7e217c3bfd7ba64 : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/afac62dfe7fc4766bee86138383248fa : SUCCESS in 23m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52b035351c134db6bd3821c8d5fb123c : FAILURE in 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3f6843c926c48ff2975ea7e7b248d4debbef62f9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-28 21:55:09.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/84007dc095c74f1680c2977eb85ac4e3 : SUCCESS in 51m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c49e649432b54155b4f5a20a1f8a629f : RETRY_LIMIT in 2h 33m 08s","accounts_in_message":[],"_revision_number":7},{"id":"5ff0f490e42d4ba223422b1b20884e236396424a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 22:08:16.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c26c742e56324b07b28ecb7b6c5bfd0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5637feb68a44991a7f19e5076e287af : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b72e9873232a4991bd9c4472d6ab307e : SUCCESS in 12m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4b3b72a774f413594acaf8a52cab6d0 : SUCCESS in 12m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/964b8f56de87471a9c5b4ff22ef4b4ba : SUCCESS in 14m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69bc9d9b27d3449a85d5fb431bfbf2ad : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/02be06425dad493bb0e02d2633316626 : SUCCESS in 1h 00m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66be3c9581a3474ea3c34cfd9de7fb6b : SUCCESS in 1h 00m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1580f98ee854c11a8989cfa862a54b5 : FAILURE in 1h 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f9f9ba2cd854fe984160d031d0d8c0e : SUCCESS in 1h 04m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f2f3de80cf54442db5d0f7c33640df8e : SUCCESS in 25m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/27d477bd5ae74828836f872610d0117e : SUCCESS in 6m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6330a56d65414450826eb77532e0b3fc : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa5379b094f740c79df7617380036cea : SUCCESS in 6m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/292883591abf4745adcd81bace8cf568 : SUCCESS in 11m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca76ae103f234e54a7aa7e43fb48c472 : FAILURE in 35m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0304aa2e60f04f06bdfe4cb9b04e9908 : SUCCESS in 7m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8269de6c320742e1bdf110fb5594b1e4 : SUCCESS in 1h 58m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f11f1fd7b1034204b48f6ae3a93c21f3 : FAILURE in 2h 52m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/46e6c2f20b0b495eb111f722eb2b0bce : SUCCESS in 1h 44m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c17b51ba88a44e00abea131167ed90da : FAILURE in 16m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/39fe0299539449ac98de5b776c17aaed : SUCCESS in 34m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7de51549e22444abfd683351ebe7d6f : SUCCESS in 1h 13m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ee8f2e654f034ce29f28d8db37b4bd62 : SUCCESS in 1h 44m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dd868e66e1a945ecbbbb319460cc9e2d : SUCCESS in 36m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c32b25382b1b4c3d984251f2bfef1035 : POST_FAILURE in 2h 02m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8fb76cfbcdd048fe8f452e2f3a96e6d2 : FAILURE in 15m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9cbfd09466e411084f175458ac6e1f2 : SUCCESS in 1h 39m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8309ae0b144c44d8a3acce29ee95200a : SUCCESS in 1h 29m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5600b1f203c64a618124240a7f23fbfc : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":7},{"id":"543ebd84b4883eee42ac0a8c31b5aa5a6bf92e9b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-29 11:14:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d764355ee273fd15c39faa6e638aa7bf52390ed5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 11:36:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d62b115ae414774889ac1010d603899\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c31264bccc6c42e985c2c746ecfb0d03 : SUCCESS in 16m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fd38021767d41af82f8bd96f601f826 : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f92837545dc4d35a4d918194e8200df : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4bbe4aaf5cf907d8a48ae9156dea2f27c2adafde","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-29 12:16:00.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/13cdef7d80a44c5986bf8bbc497567ee : SUCCESS in 45m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/289f62bf809447ad84995c8a156d0bd7 : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":7},{"id":"8e4d090ab8382f28641be23cb5fcd4cb6a3d49f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 13:16:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b331a34c131487396ef2d7b97b9e2d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d45d08f94cc476fa1b67740bc43112a : SUCCESS in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b80ecb35b97c4382a2ceb07429a57418 : SUCCESS in 10m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6eb78a86b134d5690fdf88c241a8ecb : SUCCESS in 8m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a15cc3d5e874ed595a83e54fa37c49f : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e6b5112aa44174a4314d57dc2ebe78 : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/6a53e50771b442a89b141bcfaac1c1e1 : SUCCESS in 58m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61cc76d355d048779729be870ef2a8e0 : SUCCESS in 1h 08m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13bf8954023848598861849c994c3cc3 : SUCCESS in 1h 18m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86a0760a21a8405aacf71bfc99853f4f : SUCCESS in 48m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/50cb4568248f48adad8102f8e81be3c0 : SUCCESS in 22m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0bdd0d15c0b04e0d809d1be91bf42fc4 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f07081c815674d7f9a4a91118c9d41d5 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d41506bd7a8248cfbfbf7dd9d2c25dd7 : SUCCESS in 5m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d5564f54529c4365bfc23091daf696fc : SUCCESS in 10m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/23fe34517d394a60bf6cb9d65b73d6d7 : FAILURE in 27m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2810ef541494b35827f4b8ea61dc5e0 : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70b09b27829b46d68335a55a45b2c235 : SUCCESS in 1h 30m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a0683a82381944db99bb8e0d50ee41b7 : FAILURE in 1h 58m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4675db17b0c04ff7993b253442fdcaba : SUCCESS in 1h 56m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/78765742327d451a8b82136a227242a9 : FAILURE in 17m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cded903a8b304a48aed11dda3f67e882 : SUCCESS in 32m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/90a237486db34f62a5e371991fbd5831 : SUCCESS in 1h 12m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5e14ffd896c048989c6668dce1d3cb30 : SUCCESS in 1h 32m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d0ec52f82d65433a884e7368db242fba : SUCCESS in 44m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/645eddbbea1f4116a0a62ec637a25072 : SUCCESS in 1h 24m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5ce3ede59bf54f4c8ea36db49284b98d : FAILURE in 16m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34be7f4dbb0d4b47a3ad45a9288a0cfa : SUCCESS in 1h 38m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8720cd9a35b844078d6b70dd9aaa3439 : SUCCESS in 1h 28m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72506ab0c01c4cf9b72d0100709ac6c5 : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":7},{"id":"7a079e1be8a267197c03520516edd3667e47e9f1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-29 13:17:48.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"ae3a905335ff2c4ede8abe098a7f2e78a06679cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-29 13:22:31.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_07_2025_13_17_02\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"b1ffe10fb4cd5152f44ad844d8eae6de074a3e55","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-29 14:05:15.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9ee7f46b689a61b893ad8020160d91473d9c287a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-29 14:07:00.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"7b31075d5d13dca2907e538937b4b11e0beb1821","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-29 14:14:39.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"859559024634ab1e93b3cd7af81b19a37901e445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 14:15:07.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9008e9738a026132ae8e3b3889d7be7249384ae1","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-07-29 14:15:08.000000000","message":"Patch Set 7:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2c2cf6e633e3d0fd67390975f6bc0a84a1512550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 16:26:57.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0e8b933362a4e0fa87ccfaa4e6ed56a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8fad6605e92432b80a6f0dd4554cc54 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29b6bfe03a794719a1405ab74a3f7cb4 : SUCCESS in 8m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11b83083d51149d9a9afce8eb03921d7 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8da903f92eb481d812375359ade795b : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/1de9d1d620984422bdbf3362d21fbac3 : SUCCESS in 41m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ce5b98e1f5c142a7a3ff6163ac415043 : SUCCESS in 58m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e303eab502c4768b768f662cf7b560f : TIMED_OUT in 2h 05m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b33dd39394a448e8bce83e51a34ce6cc : SUCCESS in 1h 07m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf8edd0af44144b0b8e9f64d3df954c2 : SUCCESS in 1h 12m 43s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/803c3bb2f1e042c09277b3cefadf154f : SUCCESS in 2h 10m 37s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1824b7ada8bf4a3190686fabb28b5913 : SUCCESS in 1h 08m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eb0ee05959d451b81335df29be338a7 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":7},{"id":"f1aaa95281d45a2e5264961da34bbbc53ee827c1","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-07-29 17:45:25.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/c265be700fa0437ebb658d0ac79cc741\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/e54b08af575345c78f4611a09ed46c85 : FAILURE in 3h 30m 10s","accounts_in_message":[],"_revision_number":7},{"id":"aa365c59b4a31faec5bfc59f2d42e540f35f654b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-29 22:01:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3189af2e41ec32b482bfb133eafed311019275e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 22:30:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61cbcd3d5f0e4056b46c8ec03450b20f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a013ee1ed6d44b08b13a22c54845d79f : SUCCESS in 16m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f9ddc42d67946e49a017d4d07a0d776 : SUCCESS in 19m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01ee23d03f79498084a0201aa373e08d : SUCCESS in 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2e90a720be5d9775f4eb96849803fa37fc09fd4a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-29 22:56:13.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/60fe5252cb334ee485d1db7384347a13 : SUCCESS in 51m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4c4a6aef30fe42d19931f2cc07fd9d28 : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":7},{"id":"20cef5d7af63e5a5a80c76bf7def5e34a6c0b9ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 00:02:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88db4ba2d62e4c1baa2adfc6de8b8fd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/398e96ad3a774fe282d06025fea030ef : SUCCESS in 7m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/529fadc367b2483caaa97713e8223b69 : SUCCESS in 7m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdd086d2499148c8b7a6cb0a77ab68a9 : SUCCESS in 6m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04260ef917ce4859bde541999c81bcd2 : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5fbea9d3a004c9aaa0f2a0f95b960e0 : SUCCESS in 15m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/dadebd76109641b8b28b71ab8113f5d3 : SUCCESS in 28m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0eca0736c1734ee593951b621a695bb1 : SUCCESS in 33m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b9834607c3d7478abaf9e760c327b8a4 : SUCCESS in 1h 20m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b70a61944f0740b98dafd4b5a05fc28c : SUCCESS in 33m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d40f3c96a9ea4f78ba69d1114b1a7ac5 : SUCCESS in 23m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2bcc3f8c98b244028616234e0c957b4d : SUCCESS in 7m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6cda8853a97142c285541dbe5166f37c : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bb5a2d00967941eeb4388243fb769fac : SUCCESS in 3m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/041125e712284afc8d33e86a2cdc9d4b : SUCCESS in 5m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/550c927ed2e54fce9a49c669ee9f5002 : FAILURE in 18m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/962670fa5b6b4235af558c67c4905ef6 : SUCCESS in 8m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f69f9e883839447eac05d7a11d455a86 : SUCCESS in 1h 04m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/88e6a555d8e740cfab0e1cab6c8435b3 : FAILURE in 1h 25m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d362a986ba324a398d7ecf0e47a79685 : SUCCESS in 1h 58m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0a928d7d027c426390fbd976fad1732d : FAILURE in 9m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1972d569102947f2a679959b9b83955e : SUCCESS in 35m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b7ab884b6584eccad857b2042cd44a6 : SUCCESS in 1h 06m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b7a4d999935a47658ce746fa166fc790 : SUCCESS in 1h 29m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c1acf5b08c8a4695ad76aa00f9b7a245 : SUCCESS in 46m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a4f633fd8d8c4b5b8d0ee746c95fd971 : SUCCESS in 1h 21m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3af29879436042cd81badccd8dc10e18 : FAILURE in 8m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4da69e910b284ad394e206d8c9aa2eda : SUCCESS in 53m 57s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f29674cd4fe84a0c851d5fbe2052f87f : SUCCESS in 1h 25m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62a5638ef0f54d6ab0f2d4c694c1fdc9 : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":7},{"id":"b8f5a451f743098abe3d136ceda284af0bc63deb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 00:03:02.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"372e74ac0fb2d7bd84af486f8d9620de8c45ce7d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-30 00:03:55.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"3f7e515f9e1900a9f371c43bb6c8590864bf0030","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-30 00:08:23.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_07_2025_00_02_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"da03e76aaf67729f8a592e3364c1575691302061","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 01:36:45.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddf30e5645a64dc1a7035beb4a27bc3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a05f1d6acf8f4e7db9e7251381fb6002 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/424a2eac17d9447382c3abd6f9c37e7d : SUCCESS in 9m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb9fe8e922b341af823f64ea7a83cd4a : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d1388ca0ac4335b13bd8180dd56067 : SUCCESS in 15m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/155c946ca84c4076966e5ef0227b47c9 : SUCCESS in 59m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbab8a137e88432ab591ebbe840f67c5 : SUCCESS in 34m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3b87b9cfa62436ea4c3700606d77ea1 : SUCCESS in 1h 22m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60d96090ef094017b30e513041c9c3c1 : SUCCESS in 59m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c324225462a466da1d5916427558684 : SUCCESS in 34m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ce9d440075347dba96e592a93f330c8 : SUCCESS in 1h 27m 04s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a069c9d211c4b3c9a566e875f8f33d5 : SUCCESS in 1h 29m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd6d9f0aba804edd9027e8609de907e3 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":7},{"id":"0bdbe5403fd36e3d945728af968af08526d9cfa5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 01:36:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"81c49a45294042ed8022f63b9f5006299dcb81d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 01:37:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/037610b3dd75443985711b151c8454bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847bdf89adf3418380e262e20be76249 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":7},{"id":"14bcf6d341ae47ea13e7fa2bf3f990dcf0cb53f6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-30 13:10:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1612e510e2e0954b13b49b5a6b8ed10b71f29868","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-30 13:31:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f5705689495b60df19d98d0e8d7d7a0b9a9813a6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-30 13:32:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a39fa8fb2876a98b83570295504c0f5e46da7d6a","revisions":{"2b2f0b84c004f15d3b7ccb1df0ed71da1006bb80":{"kind":"REWORK","_number":1,"created":"2024-02-20 09:07:55.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/13/909513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/13/909513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/13/909513/1"}}},"commit":{"parents":[{"commit":"7245ec027b1902dc24c30cf3885a5dfec1866a6a","subject":"Merge \"Remove six from Infortrend driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7245ec027b1902dc24c30cf3885a5dfec1866a6a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"subject":"WIP: Test glance new location APIs","message":"WIP: Test glance new location APIs\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/890509\nChange-Id: Ib70219a9d085257c90a75ddcfcb935b4659fd28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b2f0b84c004f15d3b7ccb1df0ed71da1006bb80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b2f0b84c004f15d3b7ccb1df0ed71da1006bb80"}]},"branch":"refs/heads/master"},"c02bbfcd66c792fa954c50c4307aeb03b50e138d":{"kind":"REWORK","_number":2,"created":"2024-06-24 15:22:36.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/13/909513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/13/909513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/13/909513/2"}}},"commit":{"parents":[{"commit":"7245ec027b1902dc24c30cf3885a5dfec1866a6a","subject":"Merge \"Remove six from Infortrend driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7245ec027b1902dc24c30cf3885a5dfec1866a6a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-06-24 15:22:21.000000000","tz":330},"subject":"Add support for glance new location APIs","message":"Add support for glance new location APIs\n\nThis patch adds the call for \u0027add_image_location\u0027 method which\ntriggers the new location API workflow that is more secure and\nrobust than the old location workflow.\nThis call me made when glance is using cinder as a backend and\nwe want to perform an optimized \"upload volume to image\" operation.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/890509\nChange-Id: Ib70219a9d085257c90a75ddcfcb935b4659fd28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c02bbfcd66c792fa954c50c4307aeb03b50e138d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c02bbfcd66c792fa954c50c4307aeb03b50e138d"}]},"branch":"refs/heads/master"},"e1d0711de253be9391994eb4b64f1ad2916c5023":{"kind":"REWORK","_number":3,"created":"2025-04-28 19:18:47.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/13/909513/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/13/909513/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/13/909513/3"}}},"commit":{"parents":[{"commit":"7245ec027b1902dc24c30cf3885a5dfec1866a6a","subject":"Merge \"Remove six from Infortrend driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7245ec027b1902dc24c30cf3885a5dfec1866a6a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-04-28 19:18:43.000000000","tz":330},"subject":"Add support for glance new location APIs","message":"Add support for glance new location APIs\n\nThis patch adds the call for ``add_image_location`` API\nwhich triggers the new location API workflow in glance\nthat addresses OSSN-0065. It is more secure and robust\ncompared to the old location workflow.\nThis call will be made when glance is using cinder as\na backend and we want to perform an optimized\n``upload volume to image`` operation.\n\nChange-Id: Ib70219a9d085257c90a75ddcfcb935b4659fd28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1d0711de253be9391994eb4b64f1ad2916c5023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1d0711de253be9391994eb4b64f1ad2916c5023"}]},"branch":"refs/heads/master"},"09ef775a70170d77729bec7f8bdbfcdd3a14aa45":{"kind":"REWORK","_number":4,"created":"2025-04-28 19:25:59.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/13/909513/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/13/909513/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/13/909513/4"}}},"commit":{"parents":[{"commit":"7245ec027b1902dc24c30cf3885a5dfec1866a6a","subject":"Merge \"Remove six from Infortrend driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7245ec027b1902dc24c30cf3885a5dfec1866a6a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-04-28 19:25:48.000000000","tz":330},"subject":"Add support for glance new location APIs","message":"Add support for glance new location APIs\n\nThis patch adds the call for ``add_image_location`` API\nwhich triggers the new location API workflow in glance\nthat addresses OSSN-0065. It is more secure and robust\ncompared to the old location workflow.\nThis call will be made when glance is using cinder as\na backend and we want to perform an optimized\n``upload volume to image`` operation.\n\nChange-Id: Ib70219a9d085257c90a75ddcfcb935b4659fd28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09ef775a70170d77729bec7f8bdbfcdd3a14aa45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09ef775a70170d77729bec7f8bdbfcdd3a14aa45"}]},"branch":"refs/heads/master"},"ffd93a6d10c91482d0dfd9a82b63d140affb5ea3":{"kind":"REWORK","_number":5,"created":"2025-05-06 09:56:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/13/909513/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/13/909513/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/13/909513/5"}}},"commit":{"parents":[{"commit":"7245ec027b1902dc24c30cf3885a5dfec1866a6a","subject":"Merge \"Remove six from Infortrend driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7245ec027b1902dc24c30cf3885a5dfec1866a6a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-06 08:53:03.000000000","tz":330},"subject":"Add support for glance new location APIs","message":"Add support for glance new location APIs\n\nThis patch adds the call for ``add_image_location`` API\nwhich triggers the new location API workflow in glance\nthat addresses OSSN-0065. It is more secure and robust\ncompared to the old location workflow.\nThis call will be made when glance is using cinder as\na backend and we want to perform an optimized\n``upload volume to image`` operation.\n\nChange-Id: Ib70219a9d085257c90a75ddcfcb935b4659fd28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ffd93a6d10c91482d0dfd9a82b63d140affb5ea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ffd93a6d10c91482d0dfd9a82b63d140affb5ea3"}]},"branch":"refs/heads/master"},"84a7e94cb6c7a62659c50b75a32e6d279cb982fb":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2025-07-23 18:33:55.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/13/909513/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/13/909513/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/13/909513/6"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-07-23 18:32:41.000000000","tz":0},"subject":"Add support for glance new location APIs","message":"Add support for glance new location APIs\n\nThis patch adds the call for ``add_image_location`` API\nwhich triggers the new location API workflow in glance\nthat addresses OSSN-0065. It is more secure and robust\ncompared to the old location workflow.\nThis call will be made when glance is using cinder as\na backend and we want to perform an optimized\n``upload volume to image`` operation.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/950964\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: Ib70219a9d085257c90a75ddcfcb935b4659fd28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84a7e94cb6c7a62659c50b75a32e6d279cb982fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84a7e94cb6c7a62659c50b75a32e6d279cb982fb"}]},"branch":"refs/heads/master"},"a39fa8fb2876a98b83570295504c0f5e46da7d6a":{"kind":"REWORK","_number":7,"created":"2025-07-28 14:26:43.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/13/909513/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/13/909513/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/13/909513/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/13/909513/7"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-07-28 14:25:18.000000000","tz":0},"subject":"Add support for glance new location APIs","message":"Add support for glance new location APIs\n\nThis patch adds the call for ``add_image_location`` API\nwhich triggers the new location API workflow in glance\nthat addresses OSSN-0065. It is more secure and robust\ncompared to the old location workflow.\nThis call will be made when glance is using cinder as\na backend and we want to perform an optimized\n``upload volume to image`` operation.\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: Ib70219a9d085257c90a75ddcfcb935b4659fd28d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a39fa8fb2876a98b83570295504c0f5e46da7d6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a39fa8fb2876a98b83570295504c0f5e46da7d6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
