)]}'
{"id":"openstack%2Fcinder~909689","triplet_id":"openstack%2Fcinder~master~I3e90a00707a83764de3f39986fe3db1630906ae2","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-02-21 18:17:10.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I3e90a00707a83764de3f39986fe3db1630906ae2","subject":"WIP: Expose bug 2009483 with mypy","status":"NEW","created":"2024-02-21 16:09:13.000000000","updated":"2024-02-24 09:33:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1294b2e1aac238fe0c8c99e91abb7588d62e17a3","_number":909689,"virtual_id_number":909689,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-21 18:17:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 16:09:20.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 16:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 18:17:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-21 18:55:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 19:13:02.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 22:29:16.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-21 23:51:50.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 00:04:36.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-22 00:29:31.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-22 02:47:37.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 03:41:47.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-23 03:00:33.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-23 11:45:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-23 15:53:26.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-02-23 15:55:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"e230f82a52964cd07eec585ca933b997fb449ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-21 16:09:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52cb54b9dfa1110fe36b20cbc1dbfef7b20cbd66","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-21 16:09:20.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16545/","accounts_in_message":[],"_revision_number":1},{"id":"e2f4422d3bbc6c748950d49f83066cfe7513b393","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-21 16:09:22.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16545/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cad7546634523c21dfd09ea83c209a104df81774","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 16:27:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e28ab3558323428d994d48c7bd93ea13\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87254d2fb5a54708b9e1011b50d0d168 : FAILURE in 4m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c110121b50c040cbbfa3181d52d1974d : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52de46b8eca746cda65b7f03c353e2a9 : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"849ec55dea544fc8ed02a1603edc3165d739dc3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:17:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7cf3eb738bd4c8abab0c4b66294b7b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dea939c7d584166ac667076d0113f7b : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c627f57256ea4c6ebc71761f7d230a97 : FAILURE in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f436a19750c44e6aa79c9863d82372e4 : FAILURE in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d6c7de1c824c268adfdb2bda436963 : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/5da241f31f484e489b8bce57319a70d9 : SUCCESS in 1h 01m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d2bc48ed32b43b69d7130a785c305b5 : FAILURE in 29m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f17af8f2fc54fe0b6a855bc9c9ba8b1 : SUCCESS in 1h 25m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a43e528c1ca41578fcc7fded0d722f8 : SUCCESS in 1h 01m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/affa370f8fb041c0b4742b9ef5654f45 : SUCCESS in 15m 55s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2ac0be57830d4d4488b965cd4da39d5f : SUCCESS in 12m 25s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ec069d1106394499ab40ce1f229ad04a : FAILURE in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e94267c692e84fe389792d3615e42a37 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62c16d8d4d4946318549ab8cbd7d6b44 : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d221944e2ef848498577cb336f9ec4fc : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7716ef3d87214478bde0d55d1f131780 : SUCCESS in 38m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b840e1a0e874a928030672e3c600c0d : SUCCESS in 7m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/731a0a08dcd3434384c5a9ae035c7434 : SUCCESS in 1h 55m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fa5a75a5b9554a3fa7abffd17dec9984 : SUCCESS in 1h 58m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c328fa13033440d282e779112be57dde : SUCCESS in 1h 51m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1fd327b8a4ac49c1a2cc44264917b45e : SUCCESS in 1h 45m 52s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7b5ef892de8e46ad800e0ab06be1b399 : SUCCESS in 1h 16m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6d9d54f2a60f430f9003882a56e693fe : FAILURE in 1h 41m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6628a4304513404b9792577c026efecb : SUCCESS in 27m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dd3277376ee444db97e071835230b8c0 : SUCCESS in 1h 24m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/54643e354aeb4fc7a76d99780643a8df : SUCCESS in 1h 33m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/818683da532d452e8ef1c97dc49f8a38 : FAILURE in 1h 42m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dde6d5b046654f919fcd5ecfa2300851 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":1},{"id":"20b80d27d583d7c7df24777c282a5b0d886b5d92","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-21 18:55:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909689 : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":1},{"id":"77cd9296e13c50be2583c9e83c167070bc15c5e9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-21 19:13:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/89/909689/1/ : SUCCESS in 1h 36m 33s","accounts_in_message":[],"_revision_number":1},{"id":"dca91ffc376ae0a223df7dc62c83563334eaa7be","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-21 22:29:16.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/812480371b19446e9ed27fe0f8fd020d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/15c3d04ef4324fc5bdf598719cdd93f3 : FAILURE in 1h 35m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dc42c041e25ec52c7d56d631170f031c0f657fa3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-21 23:51:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/15a02e9d60264d66bb16766b3af6a918 : FAILURE in 1h 06m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4eba77fb096b4622a5683535d86626ed : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8538291a7e76893f8c0da5d12eb8fe31acbd3c5c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-22 00:04:36.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/909689/1/fc : SUCCESS in 0h 50m 57s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/909689/1/iscsi : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9332e4bd78180187e01b87b887cb0dc95ace8c50","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-22 00:29:31.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/766/.","accounts_in_message":[],"_revision_number":1},{"id":"99d7ebe7a294eb0b0013e4ba6a5f536ec8d53693","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-22 02:47:37.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/89/909689/1/check/DellEMC_PowerFlex_3.6/ced1c29/DellEMC_PowerFlex_3.6/1510 : SUCCESS in 1h 43m 33s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/89/909689/1/check/DellEMC_PowerFlex_4.5/5293010/DellEMC_PowerFlex_4.5/474 : SUCCESS in 1h 47m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"480e632a55ee5f890c473f4e1b1cab7fc9753dd0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-22 03:41:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909689 : FAILURE in 1h 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b8d2c9ea984767cd3e40d36570c180b723987f01","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-22 06:46:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909689 : TIMED_OUT in 3h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"942867cf50c827490d89a093897871af4128d1ce","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-22 06:59:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/755/.","accounts_in_message":[],"_revision_number":1},{"id":"07d55ad507a27e1ac78a7aee0ab0c203dc2098a8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-23 03:00:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/909689/1/20240223085442/NEC-V-FC : SUCCESS in 2h 46m 12s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/909689/1/20240223085442/NEC-V-ISCSI : SUCCESS in 3h 05m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6ea104b634bd755ec05ba216b9bbd90efd5330ca","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-23 11:45:44.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/909689/1/check/DellEMC_Unity_FC/0f7a4e8/DellEMC_Unity_FC/1377 : SUCCESS in 1h 02m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/909689/1/check/DellEMC_Unity_iSCSI/d404e30/DellEMC_Unity_iSCSI/1408 : SUCCESS in 1h 36m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a4d6ced4ca6cc62e24d3e25aa8d8f44a5104ff48","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-23 15:55:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/909689/1/20240223182021/NEC-FC : SUCCESS in 3h 19m 21s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/909689/1/20240223182021/NEC-ISCSI : SUCCESS in 6h 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1294b2e1aac238fe0c8c99e91abb7588d62e17a3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-24 09:33:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/59a877b860f143d4916f669a050d82f8 : FAILURE in 1h 17m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c3a0f3b9fe934484ad1ec481c8adfa96 : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e399fe0d62e07fb4e57935056d2fe612950314c","revisions":{"5e399fe0d62e07fb4e57935056d2fe612950314c":{"kind":"REWORK","_number":1,"created":"2024-02-21 16:09:13.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/909689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/909689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/909689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/909689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/909689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/909689/1"}}},"commit":{"parents":[{"commit":"4e07b37a792d36561a6881360ddff7b748a3a7e4","subject":"Provide message info in continue_backup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e07b37a792d36561a6881360ddff7b748a3a7e4"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-21 16:07:22.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-21 16:08:38.000000000","tz":-300},"subject":"WIP: Expose bug 2009483 with mypy","message":"WIP: Expose bug 2009483 with mypy\n\nThis is the beginning of how to use mypy\nto fix the root cause of bug 2009483, which is that\nthe RequestContext object\u0027s message fields might\nbe None when they are assumed to contain message_fields.\n\nChange-Id: I3e90a00707a83764de3f39986fe3db1630906ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e399fe0d62e07fb4e57935056d2fe612950314c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e399fe0d62e07fb4e57935056d2fe612950314c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
