)]}'
{"id":"openstack%2Fcinder~909847","triplet_id":"openstack%2Fcinder~master~I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc","project":"openstack/cinder","branch":"master","topic":"new-location-api-support","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-07-25 15:37:16.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-07-30 02:43:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc","subject":"Add testing for optimized volume upload","status":"MERGED","created":"2024-02-22 09:12:43.000000000","updated":"2025-07-30 02:46:06.000000000","submitted":"2025-07-30 02:43:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909847-new-location-api-support","meta_rev_id":"7b276aab6bb759b2aa71dea4afdfad19a954d777","_number":909847,"virtual_id_number":909847,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-30 02:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-07-25 15:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-07-28 19:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-07-28 19:10:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-07-25 15:37:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 09:12:50.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 09:34:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 10:17:21.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 10:35:56.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-22 10:54:06.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 11:05:30.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-22 11:11:09.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 11:26:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 11:30:26.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 12:00:39.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-22 12:26:33.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 12:55:16.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-22 14:00:12.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-22 16:09:38.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-22 20:08:11.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-02-23 15:48:28.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-23 18:48:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-24 14:24:53.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-27 08:57:37.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-02-28 05:27:31.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-08-19 22:58:50.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-19 22:59:05.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-08-19 23:00:14.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-21 11:21:22.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-22 09:46:27.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-25 21:28:33.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-05-17 23:55:57.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-05-18 00:13:13.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-05-18 00:46:40.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-07-23 21:33:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-28 19:10:51.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-07-29 14:15:10.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"329501b92f9ca4aea4f21016250160d81437ed80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-22 09:12:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4ba9a4b7395f74193cce96c6a4ef87fd6a2af62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 09:12:50.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16547/","accounts_in_message":[],"_revision_number":1},{"id":"e13b6a1a558425eb2ec17251dc89ab5b017615df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 09:12:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16547/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"16394f388a5a871f8b45f386910d2123458a14cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 09:34:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6edbde14f7d6404c9046e4de5f12bcb7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/afc50775e2e34031ac557b0f03fa984f : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cbdacf8a51bb4e6f9e8e0beb8b4fecbd : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cf5f589385c6464a86e719456c4ac833 : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b00e38204c864466f08026dc0afb3b6b3d4a0cea","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-22 10:17:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0abb6551e26b57f479483cadf48bbd3f1dd54223","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-22 10:35:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":1},{"id":"39d4b7b870e504b516d6c8327220751bc9a3467e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-22 10:54:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/909847/1/ : SUCCESS in 1h 33m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7b71366d5d40d3a341087d9debd8715f28966bb5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-22 11:05:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/768/.","accounts_in_message":[],"_revision_number":1},{"id":"a760dd86ea148aa07e53c4a096345bfd41ddd651","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-22 11:11:09.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/47/909847/1/check/DellEMC_PowerFlex_3.6/daa026f/DellEMC_PowerFlex_3.6/1514 : SUCCESS in 1h 37m 13s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/47/909847/1/check/DellEMC_PowerFlex_4.5/8fc28f0/DellEMC_PowerFlex_4.5/478 : SUCCESS in 1h 39m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"8cad89b0e419f7ea8939766a3d6b7eb39b8d4bc6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-22 11:20:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : FAILURE in 40m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6480a6ab6d8ab4d1c75f14c43ce95f080c6be92d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:26:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e6689d9c1474f1c98f7c897e82a3d90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d972d13d9bd844e6b311b2e2c53412cd : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b679066444b48e38c9492b6e57d6e7c : SUCCESS in 12m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a960a617f06407ca239dd4399178fd5 : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d23c5193ed44fda822bd53b40ea2bef : SUCCESS in 15m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/27013085da7c48b587dbaf46d63ca394 : FAILURE in 50m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/617750864105457a8f5acaab5e44ebca : SUCCESS in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09251cf2bb7640079d241fcc61c98585 : SUCCESS in 1h 25m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/305ee708c50a4ec3b10ca412767c1521 : SUCCESS in 1h 09m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/02c575a1c8824186a16e8abb84c0a233 : SUCCESS in 14m 46s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9ca6e440d822457ca5731dc6327eb649 : SUCCESS in 14m 55s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/99ebe267836f4a4e8746fb7bd25494a0 : SUCCESS in 6m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/357b739575ed42f394a1f5e0c9a34372 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c3032e02a464cde849ea9f7334b2e39 : SUCCESS in 5m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b948df9560204c9d8c200cccbad60a03 : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1b06293732824d338fb009194b0cae93 : SUCCESS in 44m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d99183e76f34606b4f872ddae9539a1 : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff032cf5c1844d0db644739e098a5fa7 : SUCCESS in 2h 02m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4170d3ab3613412b8263348b2d6ede0b : FAILURE in 2h 07m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf9b48f39e774cca943c2fb63ce70b4d : SUCCESS in 1h 45m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fc927728f2e94d35a41cba0a43f57a3f : SUCCESS in 2h 06m 09s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7993b4e6e11943bd9a2561bb6e55b9a2 : SUCCESS in 1h 10m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/47b813317adb4ef0864694252fefbbf5 : SUCCESS in 1h 32m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a80d38e1498741139e12e9c9e78cfed3 : SUCCESS in 35m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57d0b987532143d387cbe1f3defbe75e : SUCCESS in 1h 41m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ea9c62d365124b93a3dfa0fc2f241045 : SUCCESS in 1h 32m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbed23c68cdc42e2bb4c9bdcd72b7b31 : FAILURE in 1h 48m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05e8f578e32f461b9d5fa895d9d175bd : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dabfd38dd2020ccb31b62c68d0c3a65e41ddab1a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-22 11:30:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/47/909847/1/check/fujitsu-eternus-dx-iscsi/52232b6 : FAILURE in 2h 12m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/47/909847/1/check/fujitsu-eternus-dx-fc/e24ea23 : SUCCESS in 2h 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8486f1b3283cf0490fd5197a0a55e27e7e876ca3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-22 11:43:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/757/.","accounts_in_message":[],"_revision_number":1},{"id":"f747f4b18403a1d791570daa23eaa08bd3cb4747","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-22 12:00:39.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-47-909847-1/FC/ : SUCCESS in 2h 47m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-47-909847-1/iSCSI/ : SUCCESS in 1h 58m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"de286c17ce0123fe028a4be5f410eea989f17f94","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-22 12:26:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b6f221b794a6474aabce9b6bb2d69ce9 : SUCCESS in 1h 13m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/442464312d4149a48b73c36b26ff8774 : RETRY_LIMIT in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1e575382ac98c5475d7f635758fb62dfbde27919","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-22 12:55:16.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack/+/909841 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"1c465993dad0c8393efd4066e7f9a49db9c4a1a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-22 12:58:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ac709990713312babc024d73959c6fef0fc8aa4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 12:58:31.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16549/","accounts_in_message":[],"_revision_number":2},{"id":"c24d93e6f25505bd551e1d58b5812313e3622115","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 12:58:32.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16549/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"eebbc1a1f3080129dc7da66f5af7c54b5d134d12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 13:09:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/776b4c9bc0d34df4b8a60e0c74f7dafa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/128b9e37efc1486996edf3c1b7224b53 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/16e8188c3d01412da7b3fb5987228018 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/48b1df4da1cd43108865c90cc219ffeb : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"774d46394457ecdbc38d69981219e41a7ad8f368","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-22 14:00:12.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/1/fc : SUCCESS in 0h 34m 06s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/1/iscsi : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":1},{"id":"73f9d5924eaf42b21d18e7622ace8487543115c6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-22 14:03:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":2},{"id":"7366cb11791386c4e80e3e08b42ed0c1b5c056d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-22 14:17:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aac7e9cb35764b04aef6cb409ab2e72f : SUCCESS in 1h 17m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/782cadcef21841f68c802d796d834313 : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":2},{"id":"cceb813a63200d7f511bd0029e093a43d54f52b9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-22 14:21:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/770/.","accounts_in_message":[],"_revision_number":2},{"id":"5f6a98339bf7562c9466f7487a0d7da844f72940","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-22 14:33:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/909847/2/ : FAILURE in 1h 27m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6459f3b216ef2046ab30a21a0d32c3f241391105","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-22 15:10:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/47/909847/2/check/fujitsu-eternus-dx-iscsi/99bafc8 : FAILURE in 2h 10m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/47/909847/2/check/fujitsu-eternus-dx-fc/4722e6b : SUCCESS in 2h 06m 07s","accounts_in_message":[],"_revision_number":2},{"id":"691b8e21e918dd4af1b79883eb6acd830e9d8d90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:22:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3711990751644faeb826a68b9e2c658a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57763cccaaee421998009539912e77cc : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba0c62c544e34bc28573f409e718121b : SUCCESS in 12m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11abe0d9eb9e4438a0ad30a921700729 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af188820c79f4f6b8aed6a9158e5f1a6 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/5bc874f17dff4282bd138f35b79ffb77 : SUCCESS in 1h 00m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/de6757306e7246009697e9378d93c606 : SUCCESS in 57m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e066830c6224b20b58b6ac3c3e454b7 : SUCCESS in 1h 20m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18e06d630dee49a09e1308b16b19339e : SUCCESS in 1h 00m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/408158f1370445b6910973d8e6a2fc55 : SUCCESS in 16m 41s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4bbe10ab40654b86ad46ad530f9e603f : SUCCESS in 14m 35s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd72cee133904e81859a82c84be16b35 : SUCCESS in 6m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f5fa003c82a40f9832dc782907df682 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e082f47f16348cab6bcbe015ef52b30 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/886cb760e0354421836f3b1418934dcf : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/27209ff31ed04bb0a2b304a3c793339f : SUCCESS in 35m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/972703c4b65741449ee58d4e986700fc : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/733051c23d074bbca0769e70a0379e30 : SUCCESS in 2h 01m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/50d4a74818dc4080a3ab6d6d12c58a45 : FAILURE in 2h 22m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/da64daffcacb4e4093d0121df30e1e15 : SUCCESS in 1h 42m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5c4a0654fd794578ac506be02ea48f00 : SUCCESS in 1h 45m 25s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5abd15908c7f4321ab938cc989d13aba : SUCCESS in 1h 19m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e1504cbc60dd4a44b8b44b9408124d72 : SUCCESS in 1h 33m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/376de8ae8bdc4b1cb265960efaa08668 : SUCCESS in 35m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1f369b2a541243978c2656f74755bebc : SUCCESS in 1h 27m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f403d97c3d154cae80dad2ad6071f286 : SUCCESS in 1h 47m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/861bd6b3661a4f438a48fd7be80d1966 : SUCCESS in 1h 22m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5604789619d4b01ae6150b7b200d10b : SUCCESS in 1h 08m 44s","accounts_in_message":[],"_revision_number":2},{"id":"06240862a1c243db1526c20b7d9db11fcd021641","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-22 15:43:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/759/.","accounts_in_message":[],"_revision_number":2},{"id":"c731fda9681d3df0c7fc5685763cd733af7a0e12","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-22 15:53:52.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8be44ea1f1c944f5918c48c8180c878a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e4b89cff66704a77997d666f9ddc45ef : FAILURE in 1h 27m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ff9d33e0d023ec75c8098fe5a8da0f29dc11a720","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-22 16:09:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/2/upstream-check/cinder-tempest-plugin-ontap-fcp/6953d35/ : RETRY_LIMIT in 8m 33s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/f458857/ : SUCCESS in 43m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/2/upstream-check/cinder-tempest-plugin-ontap-nfs/314548d/ : SUCCESS in 43m 26s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/afb9bc8/ : SUCCESS in 45m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"203ee3290575ad2446bbfb8db435f0e0cf03bd31","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-22 16:12:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"29f5ad1e1c3a45dd793a5464988267bcebf9302c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-22 17:30:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":2},{"id":"fc86e7766e5f108c44dcb9f5e5541fdfc8b6679c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-22 18:39:12.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/47/909847/2/check/DellEMC_PowerFlex_3.6/91c1dee/DellEMC_PowerFlex_3.6/1518 : FAILURE in 1h 41m 16s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/47/909847/2/check/DellEMC_PowerFlex_4.5/584d557/DellEMC_PowerFlex_4.5/482 : SUCCESS in 1h 44m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"5ef8c2ef001b7bf3a991e7727bfdbb8cb07bc3ad","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-22 20:08:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/2/check/DellEMC_PowerStore_iSCSI/889e072/DellEMC_PowerStore_iSCSI/1110 : SUCCESS in 1h 16m 35s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/47/909847/2/check/DellEMC_PowerStore_FC/6a964f4/DellEMC_PowerStore_FC/1266 : SUCCESS in 1h 27m 26s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/47/909847/2/check/DellEMC_PowerStore_NFS/bc31cb8/DellEMC_PowerStore_NFS/796 : SUCCESS in 1h 11m 19s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/47/909847/2/check/DellEMC_PowerStore_NVMeTCP/a99c37f/DellEMC_PowerStore_NVMeTCP/665 : SUCCESS in 1h 49m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"a9927c54653a044b140cd6100982f1aba6be3e7d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-23 15:48:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/2/20240223212555/NEC-V-FC : SUCCESS in 3h 09m 43s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/2/20240223212555/NEC-V-ISCSI : SUCCESS in 3h 22m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1dd97fe7cf8181af5f6abe450169853713e9478c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-23 18:48:35.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/2/check/DellEMC_Unity_FC/a02fd09/DellEMC_Unity_FC/1383 : SUCCESS in 1h 07m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/2/check/DellEMC_Unity_iSCSI/7970e50/DellEMC_Unity_iSCSI/1415 : SUCCESS in 1h 21m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d9206c51337bcb0a8bd9651ac165c4942f79cca5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-24 02:25:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2542dfc13aa94fca835c7e67d65624a0 : SUCCESS in 1h 16m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f9de132344954f72910202b77fc7592a : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2f64c5716b382efc6bfda8731b25abefa510d825","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-24 04:42:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d0804e20913e4dedb364b7c053a5bd91 : SUCCESS in 1h 09m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1bb3eb4d054440caacc0164345a9e031 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fbe25e766f709ba7437a897fcce8f9f63f6b9180","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-24 14:24:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/2/20240224133614/NEC-FC : SUCCESS in 2h 47m 04s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/2/20240224133614/NEC-ISCSI : SUCCESS in 9h 48m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3f3c3d66934d20edc9df319791acccbf9f0780c2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-26 09:14:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b82c1cb885516abdd3af0fe69a8a0df7cefea69a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-26 09:15:32.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4be3b47cd09165c2e1a748ddd709072a265b946e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 09:37:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68a513d7974b4887b8831b52f565a71e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a520543b95774a18aa32e816cc3c5032 : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c0e9dbca10e41489cac15a20228aa46 : FAILURE in 19m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1f7820216855402b8673c07cde6c4e8c : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"da560224f7ac62b50492d6f071180256938cf1bb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-26 10:01:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : FAILURE in 42m 45s","accounts_in_message":[],"_revision_number":3},{"id":"f0ac89e3a25ff677758db9a04e9268d3981f65fd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-26 10:39:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/784/.","accounts_in_message":[],"_revision_number":3},{"id":"71ab33701fec3e81bad72d88e1b61bd1ac06b347","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-26 10:45:14.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_Unity_FC/584bcc1/DellEMC_Unity_FC/1390 : SUCCESS in 1h 26m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_Unity_iSCSI/2c29a63/DellEMC_Unity_iSCSI/1423 : SUCCESS in 1h 29m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ef8e3d30e8bb8dbfb9b5577352496ae3208a76a5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-26 10:47:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : SUCCESS in 1h 28m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7c8527cb374638b54a1745bf3289d32f5c545355","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-26 10:56:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/909847/3/ : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":3},{"id":"e33c50c4fc86d3415c885d67dfd689860de8cdd9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-26 10:58:51.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_PowerFlex_3.6/5d181c5/DellEMC_PowerFlex_3.6/1529 : SUCCESS in 1h 41m 11s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_PowerFlex_4.5/035e528/DellEMC_PowerFlex_4.5/493 : SUCCESS in 1h 42m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"dfc2558a3b519c7a94ddb9701633fbaf98c8991d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-26 11:05:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"52d641f8caa2d429e2ef8e2d62986d1e3a94db9b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-26 11:16:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : FAILURE in 25m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b592028494126176b436188c1ab7f957fa625709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 11:48:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c5f3fe7635d4596a479fa95c4beaa4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2293de756ad24764938cde5f107a86e8 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8969fbe03ba1403485dffc6561a6d6ac : SUCCESS in 11m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a057486f0a54b24a74232aa679f4c8e : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa9039c4aa644b7a234b76c09b0b8ed : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/329c75ce6b74442386fc23fb3ce5571e : SUCCESS in 1h 07m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f1ae46103986447fae32c14f0ca6d978 : SUCCESS in 1h 07m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c24fecc6f0ac46f292681ae2d2538f65 : SUCCESS in 1h 33m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed915311a5d54cee8c487b8ef2c03b15 : SUCCESS in 48m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/66f9f29699c8442095a171f4b999de3d : SUCCESS in 18m 49s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/95d59a4c242f4871ac3745ec0b6a59f4 : SUCCESS in 12m 38s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf6cf5cbde024d89a430c9fbf58806f8 : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3ad40fb4ada7486fbcfc5247ba58186c : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/60d7717597a6408b835438414b73fbab : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/48d71f85666e47a78ab496adcc0a5ec2 : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/838639be825740f29e3f713b1fe89f0b : SUCCESS in 46m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac230126eb3d495aae08427d4ffa4402 : SUCCESS in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a78731bef2d74d68a4951b468f2a1364 : SUCCESS in 2h 22m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4194b1c42a2e4381a761afec7dee81ef : SUCCESS in 2h 06m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9156d9ca7b284764bff9bca76c2b8633 : SUCCESS in 1h 47m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2ef00e17c10e4699a56cdd87b0324bdf : SUCCESS in 2h 06m 24s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3f8b558fa2ec470b9206b001ec60f6f0 : SUCCESS in 1h 13m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5a88693d61824bce8dd2b4063fe9cb56 : SUCCESS in 1h 37m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0504c6fbc1d744448642df811ae20491 : SUCCESS in 45m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb74c8d9e4d14fe0ad15dd164ae66ab1 : SUCCESS in 1h 26m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/61c7baf1b26740d29665d8227c5d7b54 : SUCCESS in 1h 24m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ae62ae6eb10412582457a14f1db7fdd : SUCCESS in 1h 34m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33ff22d2cee945f8abae8d3ad3f244af : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7346d4b6131cadeefb0fa55054763f2c7b57aa86","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-26 12:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/3/fc : SUCCESS in 0h 32m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/3/iscsi : SUCCESS in 0h 47m 07s","accounts_in_message":[],"_revision_number":3},{"id":"40d9929c296b706b24164cb212cfc10b35db4f2a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-26 12:17:11.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c55d45bc75ac4a4e8e126f05dd69f2f9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ea89f0eed3b94824a29b21219be37fc9 : FAILURE in 1h 25m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e39f1e9b8413b55545db09fa2bf19c3f4251c444","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-26 13:41:15.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_PowerFlex_3.6/ada5eb8/DellEMC_PowerFlex_3.6/1530 : SUCCESS in 1h 42m 38s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_PowerFlex_4.5/0374f33/DellEMC_PowerFlex_4.5/494 : FAILURE in 1h 52m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"d12cfb2f78d8f0979910597f12584abbe3c0612d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-26 13:48:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/3/upstream-check/cinder-tempest-plugin-ontap-fcp/e5fbec7/ : POST_FAILURE in 58m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/38fc350/ : SUCCESS in 1h 39m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/3/upstream-check/cinder-tempest-plugin-ontap-nfs/ebb294e/ : SUCCESS in 45m 15s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/66c73c4/ : SUCCESS in 1h 37m 32s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2a4121421133892fcab4a30c05b9a321bd29224a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-26 13:53:46.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_PowerStore_iSCSI/01859c3/DellEMC_PowerStore_iSCSI/1115 : SUCCESS in 1h 08m 58s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_PowerStore_FC/f4f33da/DellEMC_PowerStore_FC/1271 : SUCCESS in 1h 32m 14s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_PowerStore_NFS/02c5f51/DellEMC_PowerStore_NFS/801 : SUCCESS in 1h 05m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_PowerStore_NVMeTCP/f7c6781/DellEMC_PowerStore_NVMeTCP/670 : SUCCESS in 2h 05m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"ffdd5b89b6cf6ce95ab5eefbc457eba129ceda33","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-26 17:24:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/3/20240226181535/NEC-V-ISCSI : SUCCESS in 2h 43m 08s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/3/20240226181535/NEC-V-FC : SUCCESS in 8h 08m 49s","accounts_in_message":[],"_revision_number":3},{"id":"e3438866c8d5cca125f6753e1b63090df40848fa","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-26 17:24:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/3/20240226225740/NEC-ISCSI : SUCCESS in 3h 20m 44s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/3/20240226225740/NEC-FC : SUCCESS in 3h 26m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c3f249549c3361f216d9c9880725c2a91586b7f5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-27 02:31:24.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_Unity_FC/f8bd342/DellEMC_Unity_FC/1391 : SUCCESS in 1h 20m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/3/check/DellEMC_Unity_iSCSI/0b30eae/DellEMC_Unity_iSCSI/1424 : SUCCESS in 1h 29m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"87221e932cbb0d6efb109d01b31431e33a609883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-27 06:01:09.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"96b54524927acd38a3e761e7cc9c5baa06ba17d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 06:22:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db181445ad6f4220a89480ee8f8f4d15\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/488618eebff4419bb2ba90f660d145b6 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01339941bc624482b8e82d3b495b0af3 : SUCCESS in 17m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8a5d606295f847498bee09705e94aba8 : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1f8b078926243d52b57f4387f770d6e5f05c3653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 07:13:11.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/890509 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"7708a6ab3a0ef447cc7ac1462e4f0ad5e863334f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-27 07:13:24.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"faf9075c8895a0f761e7e2d179b1242cd0f7f287","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-27 07:13:25.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"e9e4465469c4c38d5167dbde0be85c14c8b322b9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-27 07:13:29.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"01c6eec52930a84c8741580c0f2ab783eb3db2d0","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-27 07:20:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"64b1c10c38d0caaa882068945c2c777528aa261b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-27 07:20:03.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/890509 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"1b6f9341e08954f1ecaaf751d9d15ad03c6f7b3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-27 07:26:30.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7d1d9c1a70320e85726d154e71617a2e91ac5037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-27 07:28:49.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"1913f93bdf2414bf3b5a417a450599bba82bfcb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 07:49:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e43e0c018ff347d5b4f94bea00d39bab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eae1a2fbac0e4b92b871c103778fddcb : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3f026b1d47c473788d8f740cee50a6f : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7f8e89d926946d9a3d62c139f9c2025 : SUCCESS in 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"22d583cef325e977deda22733fd2fb5a379902e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 09:44:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c3fd3710bca46e4bdc6079bb093775e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e276f73d07c04b57ab0c3543cc1a41c1 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/575315a7fa6949e787b53ea65519e505 : SUCCESS in 11m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/599e3d8ada1e4a9f9c763f8c8f288e05 : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed779ecc919d41669181b591df7e6a13 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/56d7f9b94db2401b85c20def136c0437 : SUCCESS in 1h 05m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/56de1065b14f47419c75eb938f90fb5c : SUCCESS in 1h 06m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/118b9d3dd578400ca7a8ba592b75664d : SUCCESS in 1h 33m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1e7d90f28ec43fe825a5379e83f7f8d : SUCCESS in 1h 03m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/06ad0c09e2ba4dbb8068adb19424a821 : SUCCESS in 12m 37s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/396af8f5dc914d4cac06c50bf9b7f6f2 : SUCCESS in 12m 10s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/941b340d6b1242e1a80ae21268e1179c : FAILURE in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ccc35db419e40059611a8ec913a771b : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e25712806ec4f68ba3c21a1bf3e9ab4 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa46a85a08864334b8ea129952f5c79b : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/230e8ea62ebc4110a6c5b07f04472c5f : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/688a0f0e42a04f349a375ddb178f4808 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/724ccd4f90bb4a8bb4279f7019f61bcc : SUCCESS in 2h 08m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7160efc411734c9fa8b6841823372a72 : SUCCESS in 2h 04m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/69f7cbc6776f4c819d3f3f138428b239 : SUCCESS in 1h 28m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a19da8a19cdd4d91828d4cd93bf394c0 : FAILURE in 13m 02s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/10a10f7567ac4244ac0fd48fbab1f861 : SUCCESS in 1h 08m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9b1ac44c574d4946a7dcd69ecc20a203 : SUCCESS in 1h 32m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/897ddd7d936b408e91acc19bb2ab2b30 : FAILURE in 36m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/90451d8e85974f3a91ebe77af06dbf90 : SUCCESS in 1h 28m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/15c5b2a10f5e45e1bb5c43d97d7ab3af : SUCCESS in 1h 36m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b447deed24cd4a48a1a95991fbf2e66c : FAILURE in 1h 41m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed9a18e6c4ce4a4dbb48a31b63cc60ce : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":6},{"id":"0b246623cfcb6318677f669dc196fbb672ae8f96","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-27 10:00:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/803/.","accounts_in_message":[],"_revision_number":6},{"id":"aedb30b55aed5a9522b2783acdc627574e52c542","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-27 12:36:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":6},{"id":"9074eb4876778eb91b45ceb31f16cbc0151e05db","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-27 13:26:05.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/909513 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"9f9b8bc46e4a799cc694e0e067ea44b315e37c77","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-27 13:26:07.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"7e51b161d113ff92cc7588d2df656877abbfbcd6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-27 13:26:09.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"46ca49ed519132086749a70a671c938aa9f29306","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-27 21:14:08.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/909847/6/ : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":6},{"id":"b2c0264d86403cf0354855733c6cfc9ee0987a1f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-28 01:44:00.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/6/check/DellEMC_Unity_FC/02239cc/DellEMC_Unity_FC/1408 : SUCCESS in 1h 10m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/6/check/DellEMC_Unity_iSCSI/c706887/DellEMC_Unity_iSCSI/1441 : SUCCESS in 1h 38m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"a40bfdb40da63318b51cb216c3c444f4ca2ab3e6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-28 02:41:34.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/882124cc54104ff1b4208315f6f1d4d6 : SUCCESS in 1h 15m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/501968a694e94088885bc0aed019df3b : RETRY_LIMIT in 2h 46m 50s","accounts_in_message":[],"_revision_number":6},{"id":"fd60baa046b74c1369415441e089603244594d14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 03:57:00.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"44b13970de4981e30962912cc92c674b7bf7cd87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:10:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7d97e0665814ddab8fabf8d5682f77c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86a2a4db3a684439b883a53881171583 : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e9beeb965c54bd18dddab0eeb015c74 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a44473b680fe403ca011902199828e3e : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"16608df5fd5b639f715b92941d685be28a9d99ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 04:32:42.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2c5a0ad29550255761028abe718514fb51ac03cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 04:32:52.000000000","message":"Patch Set 8:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16567/","accounts_in_message":[],"_revision_number":8},{"id":"d783db628e3d31e5f5e27338519503ad32822953","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 04:32:54.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16567/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"8402cc074e8e7e65e28c9b8c83c6a5b723b9b11d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-28 04:33:11.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/cinder/.zuul.yaml@master\", line 345, column 26","accounts_in_message":[],"_revision_number":8},{"id":"dd65811f3160085f66d3592e5f8fd7ec5a596b78","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-28 04:33:15.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/cinder/.zuul.yaml@master\", line 345, column 26","accounts_in_message":[],"_revision_number":8},{"id":"9f677b5bb8a0d7dff64de4c53525ff2e6b979467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:33:31.000000000","message":"Patch Set 8: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/cinder on branch master.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/cinder/.zuul.yaml@master\", line 345, column 26","accounts_in_message":[],"_revision_number":8},{"id":"ca3ae135ef5008c568b14680fd377af546d7bb8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:33:42.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/cinder on branch master.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/cinder/.zuul.yaml@master\", line 345, column 26","accounts_in_message":[],"_revision_number":8},{"id":"abb7f025a565b47149c17a32d0c8363009b6316d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-28 04:35:07.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/cinder/.zuul.yaml@master\", line 345, column 26","accounts_in_message":[],"_revision_number":8},{"id":"96794d8a25fbde55f3bf40b29b65a58e5bfb00c0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 04:35:08.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/cinder/.zuul.yaml@master\", line 345, column 26","accounts_in_message":[],"_revision_number":8},{"id":"0f14380b39a550fb7c87aad5d8f1231402772b20","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 04:36:07.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/cinder/.zuul.yaml@master\", line 345, column 26","accounts_in_message":[],"_revision_number":8},{"id":"724cefc5d684626d1b16298311ab1da7aeef6721","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 04:37:14.000000000","message":"Patch Set 8:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/cinder/.zuul.yaml@master\", line 345, column 26","accounts_in_message":[],"_revision_number":8},{"id":"de41a0fa379b389bbafce2e3d3dad3775bed2196","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-28 05:27:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/909847-2/HPMSA-iSCSI/ : SUCCESS in 1:54:35\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/909847-2/HPMSA-FC/ : SUCCESS in 1:36:56","accounts_in_message":[],"_revision_number":2},{"id":"ba4064df0608bd78c6965c6f9c3dfae0577f5bbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 05:35:21.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"584641f36265577d214654faf7521b6b706cb13a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 05:35:34.000000000","message":"Patch Set 9:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16568/","accounts_in_message":[],"_revision_number":9},{"id":"edaf06f6b02a81daa05631038cfa9ab8d91926f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 05:35:35.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16568/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"87d0912ca40338cf83a0d467e6618f0a7f64e578","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:48:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a6b7e25d3bf4290b2ff25ee5c48d080\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cde5d7be9e304049bba0acff77546e2f : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e554c1412a2b4ee887f2d207a5ff55e2 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3bd1ad688908498eafcca370bdf60808 : SUCCESS in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d2331fd5406ac081f3e00aada64765ab8c964c1b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-28 06:36:26.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : FAILURE in 31m 42s","accounts_in_message":[],"_revision_number":9},{"id":"37f8cec679fb975b917f61726ee19a946f5ad045","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 07:10:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5c397498f8a0792b4c380d8ac4cc3bac29a325b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 07:44:12.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"ec72a5359e82582e097c17d0dc01abf65464f518","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 07:44:20.000000000","message":"Patch Set 10:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16569/","accounts_in_message":[],"_revision_number":10},{"id":"4d62a04ba81c543ffee6c1e7f4eed78b741b51db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 07:44:23.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16569/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"7f112efa7ac18d432e73597b1dc52940049cbb95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 07:44:51.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/devstack for 910473,1","accounts_in_message":[],"_revision_number":10},{"id":"817c44f8d68dd75fb74676733dc4d9c1c69f2245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 07:44:53.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/devstack for 910473,1","accounts_in_message":[],"_revision_number":10},{"id":"1991fffa77db401295bc52126272035fe3299d41","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 07:45:37.000000000","message":"Topic set to new-location-api-support","accounts_in_message":[],"_revision_number":10},{"id":"4dd8090a5120eb04138e68a43a73819ee90841e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 07:46:35.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"4df350e393b078961394fe297e660fa11a150016","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-28 07:46:38.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"1da9d744637ad406c2b824b1799d790c8717fed2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 07:47:09.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/devstack for 910473,1","accounts_in_message":[],"_revision_number":10},{"id":"7d4daa791fc48c1c8ae03e6070637b6b48d904e2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 07:48:29.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"c23ed02356b40394c69a7eb87d43ef0949cb836c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 07:57:50.000000000","message":"Patch Set 11: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"3081e0fdd05271f23dce28b67bc3457d3b10b478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:12:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/556f5874b595430cb103487efbf9908b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3e31a3166a84518944cd704d4443af3 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91c15ce5e87e44ca9ba732aca47a0021 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50365b21bbdd4bcd8c724a4d9ecc27ba : SUCCESS in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"58ce1d14456d754773011e6783ec8973fd74f79d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-28 08:49:03.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : FAILURE in 46m 09s","accounts_in_message":[],"_revision_number":11},{"id":"cb1b9711b9189e33e2ef36477bba0d5ff5cd8f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:54:35.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f44a064e3681438c974a8821bbb54250\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/773b433f56534ae189d19170b4d2e7d7 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0fcfd2351094835803a35702ccff3c1 : SUCCESS in 11m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d56153ae5751470f972076202c3e92e8 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3fecc0bd0d489487773a73bb8cc890 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/01a7c14c309547a389c7cfa039d720cc : FAILURE in 49m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/10309ced42624f9e961f353aea7b60bf : FAILURE in 50m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5388026ed291499f92e3d73670fb2063 : FAILURE in 6m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e29dfee668440a6aaf8a3be491b4407 : POST_FAILURE in 16m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f317728e8bde4994b878ec845dc61f82 : SUCCESS in 15m 54s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3f2f929f7e884e8d82452bbbafb8442d : SUCCESS in 13m 06s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7fae014aa2554fefb257216cb2afc9a7 : FAILURE in 6m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6d0d147efca749c7967c3131787d6382 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/34de75e83ce54086835447a788cab405 : SUCCESS in 5m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c3ee792aca2e4b6198a54c8098497642 : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/50ce9628e00c4d33b1d9addea2b0b905 : FAILURE in 5m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ce526026bd334f2683d7384f30cddcde : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a247fdd2387342839c708ad621e49102 : FAILURE in 5m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ed3059b04bf4452e83978be32c5025f3 : FAILURE in 7m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a2bd1c729014355a059ebd5b7e73f91 : FAILURE in 5m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9fb873000de24976bab3597d24d2057c : FAILURE in 23m 24s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/76d46dd0b69d4dc8b823d4ec32e457e5 : FAILURE in 49m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a8c609ae55154456a0e9915d2a5bcdb4 : FAILURE in 4m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ae2438b4b228471ab206a2d5f7a51954 : FAILURE in 17m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/23269f14a82b480f88eb435b48d714bd : FAILURE in 24m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/317c0267751e45a9ac9aec160778482f : FAILURE in 24m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8ccb8746b7c54089bc7ebcf94014ca2f : FAILURE in 17m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6edbccaef6ec4883a6c2760e7ce5bd83 : FAILURE in 23m 32s","accounts_in_message":[],"_revision_number":11},{"id":"ecf3da2511473b7d74b30434c3cc88ce9343951d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 09:09:35.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack/+/910473 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"0bb958bd8b3342a2130acc9429a94bdfc1c29590","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 09:09:39.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"b63c146696732752df434c3822f90592c72500c4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 09:09:40.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"506e124ea8bdc8335348c6dac536988361993935","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 09:11:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"eea3ac517ca925c1a570884dde01543abfd29dd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 09:43:45.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848d82a67a3749bfa54e3f41bda9c660\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8da3ed904b04ef483c680c820220c69 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11ec1615df384053b71e5ee5571aac59 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb854a31d8a64304a63a4d9d4f1d1c2a : SUCCESS in 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1bbe3f370dbc94cdf4d02df9cc94de6711dcaff6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-28 09:45:50.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/909847/11/ : FAILURE in 54m 11s","accounts_in_message":[],"_revision_number":11},{"id":"735bbf3b218c9b86df779bb1acda9fca98c5ca9a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-28 10:20:44.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/820/.","accounts_in_message":[],"_revision_number":11},{"id":"a819cfcda5b185f4b2112dcf4daa47b10826a0e6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 11:11:19.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/cinder/+/909513, https://review.opendev.org/c/openstack/python-glanceclient/+/899453 are needed.","accounts_in_message":[],"_revision_number":11},{"id":"ae4afb436c9e08f2cf30d118c383dcf7635947bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 11:11:21.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/909513, https://review.opendev.org/899453 are needed.","accounts_in_message":[],"_revision_number":11},{"id":"ca69f6eba7049122af72950059560d1513620bc9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 11:18:06.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"094394b542eb853f1002e2a4a919dbeae337ea02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 11:41:08.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57b74fe9fa6b434184658294198fac30\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d53eed1962e24c1e9286178cf5ec8415 : SUCCESS in 16m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a583d9ac8e134547aad0a31bc40386c9 : SUCCESS in 19m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b638ac0d1cd641fa85a3efdd29eb9114 : SUCCESS in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0d3f1b83603363408ae1aa92ac4011ea4314cbf6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-28 12:24:06.000000000","message":"Patch Set 11:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/47/909847/11/check/fujitsu-eternus-dx-iscsi/87cce82 : SUCCESS in 2h 08m 08s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/47/909847/11/check/fujitsu-eternus-dx-fc/46ea992 : SUCCESS in 2h 17m 42s","accounts_in_message":[],"_revision_number":11},{"id":"79673ea54647297c988a9d61fe3782678db8e130","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-28 14:00:07.000000000","message":"Patch Set 11:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/11/fc : SUCCESS in 0h 47m 17s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/11/iscsi : SUCCESS in 0h 49m 05s","accounts_in_message":[],"_revision_number":11},{"id":"b4f31d7ba3cf8f51cf4f1d5db7434ff912bd7808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:29:48.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/038213c79c3d4acf9c6a51db637902fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c3e364c4739446e91c3e86017a16dbb : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/018708e9f8d74ea092e51729078ed72f : SUCCESS in 12m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0613d5d45a4e4cc89297f8d07b306517 : SUCCESS in 9m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b69679cb2845cb90178dcf52064f12 : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/ba10d6fa20a4480fa357012c1bd445cf : SUCCESS in 1h 17m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/33d55aab39d94f018a24e204298b95bd : SUCCESS in 1h 00m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c460d38d33d94fdb8d68d92c9565dfcd : SUCCESS in 1h 47m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8490e3981723420d8d1bc191aad1e1ff : SUCCESS in 1h 00m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/13f83c7c2b41469d85b76298fa8d48f4 : SUCCESS in 16m 59s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/53fdd2b6511144d0bbbfd4982f056252 : SUCCESS in 14m 07s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6445c05640954b92a1808a1f5279e19b : FAILURE in 7m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c38fe4323db44b24be8aab5817c0fb63 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a62fde439ee490ca540f052b196c5ba : SUCCESS in 6m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2494f7338bab4a22ab330bd7f92fe514 : SUCCESS in 6m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb7e4445fd7045019f311ef5b25587e2 : SUCCESS in 49m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca8ea53b55c2475caa83e1b82fc16af4 : SUCCESS in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a508097e87fc4830b653737b9a9c29c7 : SUCCESS in 2h 10m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4afaa44dec434863a18840fbe5ad61e9 : SUCCESS in 1h 57m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/57da1db8a82e48c995c7e8fd91f69817 : SUCCESS in 1h 47m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f2865f0502224aa582237d152b625639 : SUCCESS in 1h 54m 47s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c3723d9dc8d1489d94a7af3190fed754 : SUCCESS in 1h 00m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29fb2feea8da4bbfa9a831092a265200 : SUCCESS in 1h 14m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9d05bb1990f14804aca68fb8c5d35b03 : FAILURE in 44m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a568338a276c4cbe8bc666417c728e13 : SUCCESS in 1h 16m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/04e7ebfd87194458b5fcd274113288a7 : SUCCESS in 1h 36m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/abaf459c40774f75901dee410d5a73d1 : SUCCESS in 1h 22m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6143f3f38cd34922b753ad4fbd49e48b : SUCCESS in 47m 45s","accounts_in_message":[],"_revision_number":11},{"id":"1b400e59624f6242194c72de1346425ca7cc7e10","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 15:10:16.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack/+/910473 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"edc9f7213c0102745c0e85b4f4a3c34de566adb3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 15:10:56.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cfab7ffdcc844dc4d2c8bffb3f4fd903b271b6ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 15:43:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9032dd4ae3ef46af875040e9161d37f3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92bf20861f25441ca016a7d9604b81a9 : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f38459cfb6b41f3a37c076db58f31f6 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4d72cd3885254e3e838d1099216e9523 : SUCCESS in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2677bbefc0b15fb84c7cc1a110feb6ea6452c800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:15:15.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/909513, https://review.opendev.org/899453 are needed.","accounts_in_message":[],"_revision_number":11},{"id":"58b88ab0b3263bc6a803c880b08267bae6bb000f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 16:15:21.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/cinder/+/909513, https://review.opendev.org/c/openstack/python-glanceclient/+/899453 are needed.","accounts_in_message":[],"_revision_number":11},{"id":"3f054eb073023092d179a152d8d04fb243a5bf8d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-28 17:07:11.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/11/check/DellEMC_Unity_FC/7c1e6c1/DellEMC_Unity_FC/1419 : SUCCESS in 1h 15m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/11/check/DellEMC_Unity_iSCSI/5f2c0ef/DellEMC_Unity_iSCSI/1452 : SUCCESS in 1h 34m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"e8adbb334ef2cb7109efc3fbbcbb9186673b3f81","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 17:07:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6806c94e7f6fdd8cecd869323dd16d5ce4ed26e1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-28 17:25:55.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b535619725f146e7b58588031fd196ca : SUCCESS in 1h 09m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0bef9919b60449ec9266b13e213784c4 : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":11},{"id":"290608f51b9e24c7d5355407751d9c06fe2d024b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:27:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e79d8f8d0364688a11ffc45f58d929d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1f81e73ad184e05b758455cb1299332 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6af86941899f4c4fae7921418b62e9de : FAILURE in 16m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/11ea388cdda542d6aac6ad6724620994 : SUCCESS in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3eb81f1b1df628b5ac46cf4ae33c9b828bb1f4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:01:55.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/909513, https://review.opendev.org/906359, https://review.opendev.org/899453 are needed.","accounts_in_message":[],"_revision_number":11},{"id":"e6f7ef26d324b2e313dfeae3871ed3e06b9df7d1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 19:02:12.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/906359 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"2090eae4d52c47d62b79c4707f20efc8a23a672a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 19:07:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d579451760cfef4fdd4398d107bd33522d510297","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 19:08:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dcaebccb4830e65a4e58d0c116871f9ccaefbb15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 19:09:13.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a0847dd9e2ee248336f4ae221723a1cb80736f44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:21:45.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/909513, https://review.opendev.org/899453 are needed.","accounts_in_message":[],"_revision_number":12},{"id":"38158d9bf8b76dabe1c096bc3d6554744d7e73a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:21:57.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/909513, https://review.opendev.org/899453 are needed.","accounts_in_message":[],"_revision_number":12},{"id":"3cf7b5aaf47f1af9b4e9b0547cd39912c72c2127","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 19:22:00.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"c591e84c9372fb3b1c59db370f4313a4e45a370e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 19:22:00.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/cinder/+/909513, https://review.opendev.org/c/openstack/python-glanceclient/+/899453 are needed.","accounts_in_message":[],"_revision_number":12},{"id":"c1c2f9592e5a95dd14c9c811fd510bbacd0a36e8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 19:22:01.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"a83f99097d2a9352096fd5f9f30f51553285d80d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-28 19:22:07.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"857327d203d86751f09c812b7480799dcad5fefb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-28 19:30:52.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8528961ce7bb5cccc1c8eb49c1ff70d75669fcfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:44:10.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b898a3db1f47b5bd86b8addc46869a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2053af262c3f4f6aa2a22b184314d062 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7407a8e7913f4bcea1dfd1d4c74f4ddf : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/945e2d8e3c794b439f1c12477b55801e : SUCCESS in 11m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"45dfbdbf7bf69f06d35ad816abbbec7d3aa323d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-28 20:50:44.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/32069a9a969549388f62e1b9510cc58c : SUCCESS in 1h 21m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/65bbf84173314e04be78ab349b9f4fdd : SUCCESS in 1h 14m 17s","accounts_in_message":[],"_revision_number":12},{"id":"8654e52288d1b564efd75d3dd92898365b7854af","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-28 20:58:05.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/12/check/DellEMC_Unity_FC/99c21c0/DellEMC_Unity_FC/1421 : SUCCESS in 1h 17m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/12/check/DellEMC_Unity_iSCSI/61570a6/DellEMC_Unity_iSCSI/1454 : SUCCESS in 1h 48m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"3b4a131acede3209d460f35d053f1a00bc13f6b7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-28 21:31:51.000000000","message":"Patch Set 12:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/47/909847/12/check/fujitsu-eternus-dx-iscsi/c4bcdad : SUCCESS in 2h 10m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/47/909847/12/check/fujitsu-eternus-dx-fc/8551898 : SUCCESS in 2h 21m 57s","accounts_in_message":[],"_revision_number":12},{"id":"9ca64f95b7de570b0b76c917442b3d99b8abee95","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-28 21:46:47.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/909847/12/ : FAILURE in 50m 32s","accounts_in_message":[],"_revision_number":12},{"id":"052e5f541ce23eeb7dcc1414cc3e2a349e88f08b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 21:55:39.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/123b2dbd0f484c2f8d4cdc66bd6d073f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32b93f4935da43af8a1e3721c8acd418 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/360188ff08ca4026af043dea404cf937 : SUCCESS in 11m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee87a01d31fc4a54adb91f67457e83e9 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db030a633724438886a0938ceb78309 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/21b775c173464cb6a9e13af96c562c78 : SUCCESS in 1h 03m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/90bff131bde64f90bc5817c2c3122709 : SUCCESS in 1h 04m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3973884f72e4841b5cb071870b6014f : SUCCESS in 1h 24m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fea2e32d60234619abbf220caf95c361 : SUCCESS in 54m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ebcc2b98f0d54742841c683d57fc3be9 : SUCCESS in 17m 16s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/443c27eeb263470498866dfc28101b3b : SUCCESS in 12m 43s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aacc8ef54b774b80b05eb7b12158a272 : FAILURE in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a519248b3d4f4046afbec85027edb987 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aaaf58aee60e420cbe5efc482271ce16 : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f3fedde0156842b8a7db55ada6d76c0a : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a78e5afa2514561a6f85211383cfec1 : SUCCESS in 33m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d45fab7bf3434ac8a30a90c7896fae09 : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ea0c74754f24ce7845f61fbbb5c50a0 : SUCCESS in 1h 45m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5d9150027ff94613b0b20d2f5b9055e0 : FAILURE in 1h 47m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fe0f039b65bb4fba873cf290bbdd1321 : SUCCESS in 2h 03m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c584510589d04d838e8699233d857e5f : SUCCESS in 2h 18m 38s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c6e640ff9e3b47719e3e1fa18aff02c7 : SUCCESS in 1h 05m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5cf0ff696c594ae4847db7222c274ca2 : SUCCESS in 1h 25m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5bf5f16b16744d3c8774d339e0aa162c : FAILURE in 43m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d00da137efac465d9097fbcc202f8b1c : SUCCESS in 1h 32m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dbbc9712f2f34aa798a8163f4aac651e : SUCCESS in 1h 46m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ffee446aaf7044608fe4f16de5d8efd8 : SUCCESS in 1h 26m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b43765dd771c4537a65e3f6131ac16e8 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":12},{"id":"efd39b43fc04b13d667bb05887f29c563f870073","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-28 22:07:05.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/823/.","accounts_in_message":[],"_revision_number":12},{"id":"55ca5bedaf83150fc330a2d2544b2af2cbcde849","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-29 02:19:55.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/47/909847/12/check/DellEMC_PowerFlex_3.6/2d1c711/DellEMC_PowerFlex_3.6/1565 : SUCCESS in 1h 40m 30s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/47/909847/12/check/DellEMC_PowerFlex_4.5/f309432/DellEMC_PowerFlex_4.5/524 : FAILURE in 51m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"e813f27c098293bfeb156be7e8e387bc869482c6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-29 10:18:04.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/cinder/+/909513, https://review.opendev.org/c/openstack/python-glanceclient/+/899453 are needed.","accounts_in_message":[],"_revision_number":12},{"id":"069dbff154136f34d18584e5bab4c4c527588ca0","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-29 10:24:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/909847-3/HPMSA-iSCSI/ : SUCCESS in 1:52:37\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/909847-3/HPMSA-FC/ : SUCCESS in 1:35:49","accounts_in_message":[],"_revision_number":3},{"id":"8281ee86d96a470bc0198fd0f51043387105d9db","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-01 23:10:14.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/12/20240302033817/NEC-ISCSI : SUCCESS in 4h 21m 33s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/12/20240302033817/NEC-FC : SUCCESS in 4h 31m 53s","accounts_in_message":[],"_revision_number":12},{"id":"62cfca46eb79c8012afa8f3fa427a611bfb0c440","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-02 19:59:07.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/12/20240302135919/NEC-V-ISCSI : SUCCESS in 2h 50m 11s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/12/20240302135919/NEC-V-FC : FAILURE in 14h 59m 45s","accounts_in_message":[],"_revision_number":12},{"id":"25dcdebff73279c046462bfc7e711d4259cac62d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-04 22:08:58.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"681f6d6dfcace4f3c21fb09f61118fe6c5a41d43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 22:22:06.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c822ad51d5fa4b90a6786a5e63613bde\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/619cf8b9c5744d66a069ca22de34a725 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b67b9e59b1ad475ea3ebe04483c06021 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eaec1af001eb492282086bb5eb3b0ed5 : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d20369621e5be715f3f70aa917014e41669f98dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 00:15:48.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03b4635b31f341af9b1b68ecde9d0624\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3f504c046fc47b998762b814c2e63a3 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c81267cc01d14c2abfb239b446f2ddd9 : SUCCESS in 11m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ade285b282d4276922aac89c8eb1ca6 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24c821fcb5d40968b1a2516dbaf23b0 : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/275cf65e15d242308a13d3d7c4e329d4 : SUCCESS in 1h 02m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/25723cb32fd34801bab1712e8cc6426a : SUCCESS in 1h 00m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65fcaeea207b4ec9817ec615507d0c9b : SUCCESS in 1h 20m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d839caaeb8204277a4e0810d1e8e6658 : SUCCESS in 56m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6623c2d86bf843cb94cb5dba0238f73c : SUCCESS in 17m 44s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/61267ab94edd4612b4fdc98f7eebf3d3 : SUCCESS in 12m 45s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3df29c089ccf478880a95492e0eaa7e9 : FAILURE in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6bd5136d6ed4048a57e8ea249ea2a33 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7cc285922539424f983bd37f8e4e861b : SUCCESS in 5m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a526e57cc9294f1eb09ebe2fae436179 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1c77902ad8e42228d6065e18681e2e3 : SUCCESS in 33m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0712e19c7ece46a381aed9db62cc61eb : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b77592f9c720430fb6805811cc54238e : FAILURE in 24m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/96979f1778d74607aca79db3eabb34c6 : FAILURE in 14m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00aa4089198e4b6896ed12d5185cccdc : SUCCESS in 1h 45m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e9270e06f92843e7afd3f0ba0070cc72 : SUCCESS in 2h 00m 32s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91914a081e1e46718e50b6f18598728d : SUCCESS in 1h 10m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5ec25db0424c4867a3439ba4019f9439 : SUCCESS in 1h 22m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1b64560f8dd2465982faf1d4f756159a : FAILURE in 29m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1497d51e330c47dca30053d8fa62aceb : SUCCESS in 1h 23m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/69de7fba80ac4d8fa03e5e071fcd9927 : SUCCESS in 1h 28m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f003a33b6ab34b55a8696b8609815a9c : SUCCESS in 1h 44m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7596dea7389d435697c45125a57a2fc6 : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":13},{"id":"daf0d81366733fccb760773f78353bff6c9cd101","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-05 01:14:42.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":13},{"id":"45c969d8b877da51af566b8d6f3133bbc95cc960","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-05 01:35:49.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/06203b4f6dbe4cb596ff022e64c3a63f : SUCCESS in 1h 10m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/345d0660d4354b178637baa087c20d62 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":13},{"id":"a216e114983bf4442db50e6404502f4525315d81","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-05 02:16:49.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/47/909847/13/check/DellEMC_PowerFlex_3.6/c71580c/DellEMC_PowerFlex_3.6/1597 : ABORTED in 1h 13m 15s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/47/909847/13/check/DellEMC_PowerFlex_4.5/00fb805/DellEMC_PowerFlex_4.5/549 : FAILURE in 48m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"6b5e64f8c2814c3675e19c69d2d532e0c5b65778","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-05 02:38:10.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/13/check/DellEMC_Unity_FC/d15ffc5/DellEMC_Unity_FC/1453 : SUCCESS in 1h 20m 35s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/13/check/DellEMC_Unity_iSCSI/d57b34e/DellEMC_Unity_iSCSI/1485 : SUCCESS in 1h 36m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"6f4128b6996148e662c10baf80d483c76fb3d2fb","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-05 04:00:07.000000000","message":"Patch Set 13:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/13/fc : SUCCESS in 0h 40m 30s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/13/iscsi : SUCCESS in 0h 55m 38s","accounts_in_message":[],"_revision_number":13},{"id":"e7ed47cc3caa30a0b842d62e183703f0240f6881","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-05 04:46:38.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : SUCCESS in 1h 15m 33s","accounts_in_message":[],"_revision_number":13},{"id":"c00e25c5268dd54eb2a0bbc9b980b96df9bc45b2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-05 05:05:18.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-47-909847-13/FC/ : SUCCESS in 2h 30m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-47-909847-13/iSCSI/ : SUCCESS in 4h 30m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"6fb9b98a0270d5c74a27c9d50a8afaf8ebfed3d0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-05 05:16:51.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : FAILURE in 23m 55s","accounts_in_message":[],"_revision_number":13},{"id":"1a0ec8041326ef21c2f750ced1c4fde4fef93032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-05 07:10:32.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"46452771fbf543a11a96cad93b75dd2542fdd30d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-05 07:10:41.000000000","message":"Patch Set 14:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16589/","accounts_in_message":[],"_revision_number":14},{"id":"494a252e70737b4452a947f382eb9c8d8543f9d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-05 07:10:44.000000000","message":"Patch Set 14:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16589/ : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"b747f45ef2459d9720df950a45e6204fea5660dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 07:34:55.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60d48f13e0e5467794b4387e4467a5b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6af54000132468d844195170b8f883b : SUCCESS in 21m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/64aa6e4a3b334ab59c4b8364c2b63b05 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8728a9f6dc15405e850c5176341b1ce5 : SUCCESS in 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d1cd1484e71154a83085a22444d6873d1aa17c37","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-05 08:35:33.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9f874ea4f8a53d93a50e64c9e0476a5ba67aca7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-05 08:35:51.000000000","message":"Patch Set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"2a7c9d96432c0bdeebca35934918bdf2e5e02677","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 09:00:40.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab7d399d87e949a68a1ed87f04a07e5f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/577bbaf1b9844dad97c34a27c1d85665 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/61d710b3341b4fd7ba2ac4975bfe9370 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4bbe04361eee48b5b0039329d5e3535c : SUCCESS in 22m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"54da775023712d0b6a1468a18904378ad68c60c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-05 09:41:09.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":15},{"id":"50639d1e5c218a53335429d633e91d49346f180a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-05 09:54:50.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d9cb18bd053b4539a4b1423436dd3ec6 : SUCCESS in 1h 17m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/21928fc6236c4267965926910527404a : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":15},{"id":"2b8781cb186a1a94928aaf9a007228932aef62b3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-05 10:22:36.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/15/check/DellEMC_Unity_FC/d91bd48/DellEMC_Unity_FC/1458 : SUCCESS in 1h 18m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/15/check/DellEMC_Unity_iSCSI/2d7a6cd/DellEMC_Unity_iSCSI/1489 : SUCCESS in 1h 37m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"6da53f49ebcf787e62b2e5f68d0b33af499e71e8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-05 10:28:03.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a562aaae20b34d2db0f63baf6e3aa3ca\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1f807427c1cf4b4db0a37cfc9ea1461c : NODE_FAILURE Node request 200-0000002676 failed in 0s","accounts_in_message":[],"_revision_number":15},{"id":"faf3d89f7c2209faa2c3b2390498170e8d05f96b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 10:43:20.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a7da3ef5dd64da393073cd29562a4dd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ab2d691786401d9cbadeedaed5bf4f : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57a526798575425ebce91a6f7f0be2c5 : SUCCESS in 11m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6cec828cc394b34b76d52c373e5f81d : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7b646bd2404e8aa9f25d1e74e5334a : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f36df3a535d749748a328e4b75cb8c7d : SUCCESS in 1h 02m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4edb8e47e7964f1ca39ebb351e024243 : SUCCESS in 1h 00m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19240cf966144670b512beb80e46e7d1 : SUCCESS in 1h 26m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d670f53e76a4674856c7108800e08a4 : SUCCESS in 1h 04m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e23400a84e3642c9ba9597978cceca20 : SUCCESS in 16m 08s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/671bbcee43f146cea647a55ab8149117 : SUCCESS in 13m 42s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e896bfcea432499da5dff32fbce5b690 : FAILURE in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/38ae19c049764a9b856f878c93cc54e2 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/801c763249a24e4c81f5f7c9710f4904 : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f841ac23285147d18e3f1e2375380e89 : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/914b0b49f5444abaaa8b508c1cdf37fb : SUCCESS in 47m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1395a83d0b63413ab44eb8d3b5c3a0cd : SUCCESS in 7m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fceb3be3243c43289877e5192e457f62 : FAILURE in 31m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3fb43de4b9624e8d8c62975a94b7db27 : FAILURE in 31m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/41e0726b8c644c63bae69cfbd3e63c41 : SUCCESS in 1h 44m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/624f464aaec447d38bc6a652abc1176a : SUCCESS in 2h 01m 36s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dc85012874b34e218f175fdc01903ca6 : SUCCESS in 1h 18m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2b5e7ce7a8e14ca6b2998ea33ee39f77 : SUCCESS in 1h 38m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/50a4dc2641da433dba3e9728431769d8 : SUCCESS in 47m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/271d3b929f3146dba1dc33be5964345a : SUCCESS in 1h 28m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eb25319460ba4a318beba33549644603 : SUCCESS in 1h 33m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58efda152b5340aebb5f30c0e32d206a : SUCCESS in 1h 39m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b935fd8e736545f6a2457ca78bccfa30 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":15},{"id":"59fb2a254bb05d28889cc32e1cc2445e4bfcb847","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-05 11:00:07.000000000","message":"Patch Set 14:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/14/fc : SUCCESS in 1h 19m 09s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/14/iscsi : SUCCESS in 1h 08m 15s","accounts_in_message":[],"_revision_number":14},{"id":"fb6461f250e6490817cceadca3f00e747942d5cd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-05 11:51:02.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/47/909847/15/check/DellEMC_PowerFlex_3.6/72807c2/DellEMC_PowerFlex_3.6/1602 : SUCCESS in 1h 32m 28s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/47/909847/15/check/DellEMC_PowerFlex_4.5/2381fde/DellEMC_PowerFlex_4.5/553 : SUCCESS in 2h 26m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"c337f223ed7b4a92d664f769f07fd323cb0e545d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-05 12:00:07.000000000","message":"Patch Set 15:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/15/fc : SUCCESS in 0h 39m 02s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/15/iscsi : SUCCESS in 1h 22m 21s","accounts_in_message":[],"_revision_number":15},{"id":"72fcee36c8b9d01a3ff173e64b39ca12e097d456","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-05 12:03:07.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : SUCCESS in 1h 26m 14s","accounts_in_message":[],"_revision_number":15},{"id":"1f4c56bcdb5b0193bcfa16d516eb762476cb2266","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-05 12:20:26.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"96506f01a245bac3158e5ed1e279e948289a3c95","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-05 12:24:31.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-47-909847-15/FC/ : SUCCESS in 3h 48m 19s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-47-909847-15/iSCSI/ : SUCCESS in 2h 17m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":15},{"id":"1b61b3ecf75dbc5df1d587e1a2351a4fbef4d06d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-05 12:32:25.000000000","message":"Patch Set 15:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/47/909847/15/check/fujitsu-eternus-dx-iscsi/bdd4242 : FAILURE in 2h 59m 52s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/47/909847/15/check/fujitsu-eternus-dx-fc/64ba9e9 : FAILURE in 3h 53m 31s","accounts_in_message":[],"_revision_number":15},{"id":"9f3fa138239c0d91625f8b42a8f7881fe3cb120c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-06 04:22:40.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/15/20240306085418/NEC-ISCSI : SUCCESS in 4h 06m 11s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/15/20240306085418/NEC-FC : SUCCESS in 4h 28m 18s","accounts_in_message":[],"_revision_number":15},{"id":"c6b41b45ebf7eadb98aff2a5c32c8c7d61312c6f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-06 05:26:19.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/909847/15/ : SUCCESS in 1h 43m 17s","accounts_in_message":[],"_revision_number":15},{"id":"aa80fac1ce616510a4a07a6163ddd465c69852e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-06 06:12:35.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/15/20240306115144/NEC-V-ISCSI : SUCCESS in 3h 15m 45s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/15/20240306115144/NEC-V-FC : SUCCESS in 3h 20m 48s","accounts_in_message":[],"_revision_number":15},{"id":"4091b85c870b57f1a64ddacb1b477617f31a0dda","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-04-16 09:30:56.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"218c1682090e3ab5aa9055b035432d4e7ed961dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 09:53:22.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a44879f519774b7eb6594a663d20327c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a4b7270aa7b43f19356d3f756836b12 : SUCCESS in 19m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c875a7e3d45a4f7abed8321990fd7d95 : SUCCESS in 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c2da54755a611b11bc6bc0872f5619086532a622","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-16 11:10:49.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f92b25f370924bd1815d18f4ab8c929c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e51c870101d94473ad6de48dd43c718a : FAILURE in 1h 30m 31s","accounts_in_message":[],"_revision_number":15},{"id":"583f7f4c354c0aeda87488d9638eeb389f56af00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 11:44:28.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08f2ee822c7f4ee081d10edd5f6164ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69581287ec60423c93f41dce57ebf1c4 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80351735fa0d43bd853ad0f0343b3e95 : SUCCESS in 14m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19d52ac69d0f44d8a17a1946c4f7f935 : SUCCESS in 12m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/013627d5f4aa49aa952de1eba6aa2ac1 : FAILURE in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f53c4b0a8aa46cda1f2056991b0189b : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/3e52d568f4864ff6be7a3e1cb153b621 : SUCCESS in 1h 00m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16e75b0b8a3944ff8c0838b1703a843a : SUCCESS in 1h 41m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ada2e52180741ada617fb69597e6750 : SUCCESS in 1h 04m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/748d723408834d4895f9ff38e13b7f35 : SUCCESS in 16m 06s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/748a7f270d1144f3b48808648cc12aa8 : SUCCESS in 13m 08s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ec9f6338b63b43408d68dba7524f7f93 : FAILURE in 7m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cff92c847df64339a9a3c6d9813fc564 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c4552ba4d4f8447abaef80475f813a84 : SUCCESS in 6m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/619982bc9a76465ea71dc865c747751f : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1b2d0048cf1b4bb48f34ff1a0c4c3af7 : SUCCESS in 47m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a2099fdc2f9401390cb79485b8bbf15 : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b0c363702fa4dc7b78c1b160d91ed18 : FAILURE in 2h 11m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/78e6d97c571849d58a4b7bc8253e50d0 : FAILURE in 2h 04m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e0094e3739147758a7a379af04b74fb : SUCCESS in 1h 30m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/82e9b696e77b4e2d82a83c9a6653ffa5 : FAILURE in 1h 44m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/15dd2df39b114bd78066c7603e35edec : SUCCESS in 37m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a20fdceb8c149c3befddae48e2c2391 : SUCCESS in 1h 11m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/567fe0e80914433c965f6bb589078117 : SUCCESS in 2h 00m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b6682634e6b74d87b62a47670c142936 : SUCCESS in 33m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a422b63ab72a493ba91ecdf0dfed0e90 : SUCCESS in 1h 12m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0c0bb8e543bf43be9bb7c7e1b10afc3b : SUCCESS in 1h 29m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b9fb061d9374f2f85918be912ada81e : SUCCESS in 1h 19m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53b705a2afde4243b906f18e72ce65ed : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":15},{"id":"28e6ca3ffc8ed51c0478fc146a9d6e6cd5f83307","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-24 15:24:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4f7e91db1fd11c053438ada9837c0a51bacb7251","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-24 16:05:19.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7737f9ae5b22400b9945d3db1e9f1c4c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f63fdc2b7f97471d95f46af458ae0f9a : POST_FAILURE in 24m 45s","accounts_in_message":[],"_revision_number":15},{"id":"4259afb053dbae67ba4b3f194a4247c8868ed4e8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-25 03:26:12.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4db78d41c9e346c794338c01a6915b83 : SUCCESS in 1h 11m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eb7719715f6c4f19922f3688680d8c22 : FAILURE in 1h 16m 07s","accounts_in_message":[],"_revision_number":15},{"id":"ceacde856b3cba0da6b41f7e96386e97663c250d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-15 08:33:32.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c042ebe8b9f7d33c41998e6db1f1c5a066124aa0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-15 08:47:04.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack/+/910473 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"dd493c55431b21b9fa118ccc5e5bd19575e34d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 08:52:51.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/910473 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"cec0e1ae5f4817602e1403ade4e4fe9041738c53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 08:53:52.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/910473 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"94628b094bbcf385eff47786a481814aa6747799","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-15 09:52:07.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5ac67fce612240a9b2e4e7eb9be19ee4 : SUCCESS in 1h 16m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a98031a062d84fd791604f6bbc3c8ed1 : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":15},{"id":"e6334a28c81a8690ad06ac3c96f50339669b00e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-15 20:27:13.000000000","message":"Patch Set 16: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"c368e3c890e33db0bf93482985afdc3a224f7b7e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-15 20:30:12.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"397a15c33450015fb062f0ea42f05ce205f24f97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 20:39:27.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7af96223b75c44a4bb788ce640cc25d0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef8114c0a6b0438d82a85e8f57f864fd : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ebc0d719c1fb4b63b95a32caa2035115 : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0cbc16c60eed5c60048611ed508a126bc984f293","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-15 21:00:06.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/712a117b7fe348778c4e4c92922ae1da\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/050c2ea6af3b400ba739ef47b2b9d3f8 : POST_FAILURE in 24m 54s","accounts_in_message":[],"_revision_number":16},{"id":"11da7dd35587f9971a2ecd28318d83ad0050e326","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-15 21:18:52.000000000","message":"Patch Set 16:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/249/.","accounts_in_message":[],"_revision_number":16},{"id":"0e0078aafbee081efdd54efabef4e06eef6bd0d1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-15 21:47:49.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3e0dcec5609a4f2ab54cb162d37a99cf : SUCCESS in 1h 14m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a5c55cf33b2947d190ba78b2fd1c78b4 : SUCCESS in 1h 18m 52s","accounts_in_message":[],"_revision_number":16},{"id":"7c7d294acb6ddc84d9c16e88f3ac5e6f5b903bec","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-15 23:46:32.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/734/.","accounts_in_message":[],"_revision_number":16},{"id":"dc6df09d9322279fbd5a69591370334d2cc51bdf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-15 23:54:55.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/16/check/DellEMC_Unity_FC/ac299e5/DellEMC_Unity_FC/2233 : SUCCESS in 1h 23m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/16/check/DellEMC_Unity_iSCSI/653a9ee/DellEMC_Unity_iSCSI/2242 : SUCCESS in 1h 58m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"3e5dd0f73be11c0c71469cbf2d7a8186dccc509d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-16 00:04:46.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/47/909847/16/check/DellEMC_PowerFlex_3.6/64914d4/DellEMC_PowerFlex_3.6/1718 : SUCCESS in 1h 45m 45s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/47/909847/16/check/DellEMC_PowerFlex_4.5/3e3508c/DellEMC_PowerFlex_4.5/599 : SUCCESS in 1h 56m 31s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/47/909847/16/check/DellEMC_PowerFlex_4.x_NVMeTCP/ab5f6b2/DellEMC_PowerFlex_4.x_NVMeTCP/554 : FAILURE in 14m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"577199bd869ef36279814fd4cb68f69ea3d088e8","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-16 01:00:07.000000000","message":"Patch Set 16:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/16/fc : SUCCESS in 0h 39m 08s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/47/909847/16/iscsi : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":16},{"id":"454737746cf1e672f014eed872f3f27fc83529c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 03:37:56.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31e96883e303459aa2ac1992ad90789e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c56cefe1ab4eefa2427556ed99b57b : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0ecf466d1274d2cbdc03da8f015a533 : SUCCESS in 11m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b89ce55fcdf54277a45e53e3e4d7de49 : SUCCESS in 12m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8abd6b6de2ae4a49bb045a6278408aff : SUCCESS in 16m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333de3c2799c47ebb9f7e85496b5b967 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/dfbbdf6b9ef442d0a75f6b0d6235b98d : SUCCESS in 54m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78d98d2eeab846c79e2e84f4c22364d9 : SUCCESS in 1h 19m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3f5cd7b60814526a86a24490a788634 : FAILURE in 58m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f353b9d6f61e40aba68deeece1ae7f7c : SUCCESS in 19m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/15606f88490e4398a1d859bd74d0702b : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd65ec58ef1d4493b87a013e0ebcb5f2 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd0281403d724ea3a489fac8ad4076d1 : SUCCESS in 5m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/018b9aa36eef46bbb905c3c8d2d50fa8 : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ba2cd3c951ef4ef381a30d5446707704 : FAILURE in 36m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1efd6150c5724aada9fdd9190186e677 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f88738078f99409f8ad6fd7053a5a978 : SUCCESS in 1h 37m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/78588d565033455cbd01322b030950a5 : RETRY_LIMIT Host unreachable in 2h 17m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fd89f962defe4989b587185d3d9ed004 : SUCCESS in 1h 42m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6dc14d96787d49a9832434f094ca9518 : SUCCESS in 1h 34m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/465c80fb27854042b33ffd8835c2162d : SUCCESS in 25m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c00dc0f117184407a22c91a4a62ba662 : SUCCESS in 1h 15m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0976991bf0ba4f6f8008d4c891f30afa : SUCCESS in 1h 24m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/438e71db313b4a349f9ab2cebed6809b : SUCCESS in 37m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f9282c41ac994569b226bde5cccb5500 : SUCCESS in 1h 30m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dba41be66b3f4c77bba2000c0ab69ae4 : SUCCESS in 1h 37m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a438bc8a9baf49f9a156d64db5d3da2a : SUCCESS in 1h 36m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d249825189cf4ba6a286fdb81ed2971a : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":16},{"id":"a1257a9694c1a25572356e1efa79dff617ea0033","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-16 17:51:39.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/16/20240716225234/NEC-V-ISCSI : SUCCESS in 3h 52m 19s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/909847/16/20240716225234/NEC-V-FC : SUCCESS in 3h 59m 01s","accounts_in_message":[],"_revision_number":16},{"id":"461932db4e9271ddc10f7fa78152056fedaf5eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-19 22:58:39.000000000","message":"Patch Set 17: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"0d51247b9593de64ae6710d5ebb070569311bb07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-19 22:58:50.000000000","message":"Patch Set 17:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/324/","accounts_in_message":[],"_revision_number":17},{"id":"856c3408f7657a9efdc8c66de54d643ca463e6d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 22:59:01.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/python-glanceclient for 899453,5","accounts_in_message":[],"_revision_number":17},{"id":"2766d9dee4e6a03e46e6c73c95dcdf992f8970e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 22:59:04.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/python-glanceclient for 899453,5","accounts_in_message":[],"_revision_number":17},{"id":"dde1381929b2a60a04321d37f7c08b400baccc0c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-19 22:59:05.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev-third-party-ci/openstack/python-glanceclient for 899453,5","accounts_in_message":[],"_revision_number":17},{"id":"edc8540465647f1d14f2016d8a015b7eb594d6ae","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-19 23:00:14.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"fa73364f4e39927465223d617e20f4601dd05e19","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-19 23:00:20.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"c100eda4d41be7d4a197348f74764fd5c1c20ea7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-19 23:04:35.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/324/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/909847/17/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/909847/17/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":17},{"id":"530a6938b3591eebea91848e6fc9016327edbac9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-19 23:06:04.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"e0cdace16a6301656a27f1ab3064e0da1a770cce","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-08-20 00:29:52.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/909847/17/ : POST_FAILURE in 1h 16m 59s","accounts_in_message":[],"_revision_number":17},{"id":"3ca720d6b192ca9392b0a971a7e97462e070f24f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 01:10:30.000000000","message":"Patch Set 17:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/335/.","accounts_in_message":[],"_revision_number":17},{"id":"1eda4e971bfbd8358ebb88e61584da2102045ff8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-20 01:41:23.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1d6213c57c444859b547eab70a328ced : FAILURE in 1h 27m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5648b2ca4eb143e489831c994d3d58e0 : FAILURE in 1h 13m 12s","accounts_in_message":[],"_revision_number":17},{"id":"cb8e0599830997f6489b435f7e652a83eb44cbdb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 02:13:56.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/922/.","accounts_in_message":[],"_revision_number":17},{"id":"ce3055d4370d2ac435e49e14eca84f43acd55964","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-20 03:39:28.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/17/check/DellEMC_Unity_FC/f0c5234/DellEMC_Unity_FC/2452 : SUCCESS in 1h 44m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/17/check/DellEMC_Unity_iSCSI/4b10dda/DellEMC_Unity_iSCSI/2455 : SUCCESS in 1h 51m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"90375487cca245efdcef590213ff6c1192f38aa7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-20 20:07:16.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/17/20240821004725/NEC-V-ISCSI : SUCCESS in 4h 00m 48s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/17/20240821004725/NEC-V-FC : SUCCESS in 4h 19m 48s","accounts_in_message":[],"_revision_number":17},{"id":"e1b6a169ca8db45cb5bf7d94f61250789e5dfd23","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-21 08:18:47.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/17/20240821113224/NEC-ISCSI : SUCCESS in 5h 18m 32s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/17/20240821113224/NEC-FC : SUCCESS in 5h 46m 20s","accounts_in_message":[],"_revision_number":17},{"id":"a9f452fa64a4fe859c494ae767014ca0c2f28b2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-21 08:31:11.000000000","message":"Patch Set 18: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"308e7d429515ca73edc49cf0b5ca551ee8091ba1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 08:31:19.000000000","message":"Patch Set 18:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/339/","accounts_in_message":[],"_revision_number":18},{"id":"1601ec68a07313aaa1d5cb1c7a767017ea782510","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 08:35:06.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/339/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/909847/18/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/909847/18/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":18},{"id":"9206fd752b70ab04f3a8774ade5bc9fb9cc69a01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 08:55:53.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e19d753026c4b7da759aaf8c7c1d2f8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/264eda5989074a7187a807fb7308dbcc : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/631838cab2c549c793f9e19dee8f40f9 : SUCCESS in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"10c5300f216beb281642588969fd82966678c4d7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-21 09:20:47.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/909847 : FAILURE in 31m 29s","accounts_in_message":[],"_revision_number":18},{"id":"af5f6b975f3691e679cd10af682ff3d81317e4fa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-21 09:25:32.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : FAILURE in 32m 31s","accounts_in_message":[],"_revision_number":18},{"id":"1ac71580840d7cb6440e729baa9e9a1ea0b37ed3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-21 09:43:55.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : FAILURE in 33m 51s","accounts_in_message":[],"_revision_number":18},{"id":"6ffa55058502701c11458da6d755d29e05e23d76","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 09:58:51.000000000","message":"Patch Set 18:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/352/.","accounts_in_message":[],"_revision_number":18},{"id":"f75b37d6803b03cf78ac8e8fff39f7e86ea2b900","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-21 10:32:15.000000000","message":"Patch Set 18:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/355e75ee88e5495cacad4e82762c3f42\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/82606be0fb014e868535f0ab8bf3e07d : FAILURE in 46m 50s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bde806d9003e461db90f369437021679 : FAILURE in 1h 47m 52s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7e7ff51cc7cc48bba3b7a53e904f67b8 : RETRY_LIMIT Ansible setup timeout in 1m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a0c20185f5743b5972b53fad1048912 : FAILURE in 2h 00m 28s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained... (truncated)","accounts_in_message":[],"_revision_number":18},{"id":"d53bb7c82721993cdaff6fabe3269836bbf6c3e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 11:11:27.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49e234969004446a8e82c3b7ba5c5fd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6e884476f44b38ad28a9452b2f58b8 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccf66700e57546a6873ac8cead5fb764 : SUCCESS in 11m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/986a671d98bc445eb13d97a0c0cd85c5 : SUCCESS in 10m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19c126f10d4447478d55554fa33d7901 : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edd122329509476aba9e6fba341806d2 : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/b26be7e96325422b869daf7c3439e524 : SUCCESS in 58m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88a658b47f6e41efa2cf4723faa0f5b7 : SUCCESS in 1h 17m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9a40596400f4fc89bac32ceefdf6611 : SUCCESS in 1h 00m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd6a632b030342d091f998131e5e8a20 : SUCCESS in 16m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c57e255900bf49b295cd0b99462431d7 : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f3a92bf877044ba8b4c000645b3e6bf4 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/722835bd1be44e76ad40b150fee76ac6 : SUCCESS in 6m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f2253396ce14f18859b23d99a511928 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/565685c1a3c847818b616b7d918a7afb : FAILURE in 40m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/76ca003275174e32ab5a30374b521d4b : SUCCESS in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/55ce104c0f014e42916e3ce8224efb4d : SUCCESS in 1h 49m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/de182ab4ea69429e8be24c8005e1a104 : FAILURE in 2h 38m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb82c28819cb4ee5af9d27505c4c34e8 : SUCCESS in 1h 47m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5bc08c496c8e4a779973b92069e4d475 : SUCCESS in 2h 03m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/74dba6a0961b40f08ab6ef7ad32d5c77 : SUCCESS in 34m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7b10789d7bda444bb1696d4dcf686983 : SUCCESS in 1h 08m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/824c615c6a574bd1afc68b0f4809c047 : FAILURE in 1h 46m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/77d5918faa63440c8d6fa9a0b14017fe : SUCCESS in 46m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/98866815e2c148d48a7ddea10966a624 : SUCCESS in 1h 24m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a1e8797c009549cb81b3a5089150132a : SUCCESS in 1h 32m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72e53fa30514433b809e24ab8770c036 : SUCCESS in 1h 40m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/508c3f12e41f4efea3a143a509a3bb90 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":18},{"id":"fac19d81fb6313d34ded86bdd40dbe551e190a4e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-21 11:11:39.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3c4c6acebf10442fbe9629540e9fd801 : FAILURE in 1h 22m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/76ca6d3316c3434296838c8f1f019fe0 : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":18},{"id":"c53a6168c1964db3752a9f9394a259f651587679","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-21 11:21:22.000000000","message":"Patch Set 18:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_08_2024_11_15_48\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":18},{"id":"59b11a6b8c2dfda681490772f785c3fab5f11509","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-21 11:32:55.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/18/upstream-check/cinder-tempest-plugin-ontap-iscsi/7a8fd17/ : RETRY_LIMIT in 7m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/909847/18/upstream-check/cinder-tempest-plugin-ontap-nfs/b480e6f/ : RETRY_LIMIT in 3m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":18},{"id":"6f08e42f3b5e302e67994a3e809e6702017f54c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 11:52:21.000000000","message":"Patch Set 18:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/937/.","accounts_in_message":[],"_revision_number":18},{"id":"010db7809dab45e19ef633d30602f4ef18863861","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-21 13:21:45.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_Unity_FC/2fb92af/DellEMC_Unity_FC/2466 : SUCCESS in 1h 49m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_Unity_iSCSI/80c3bba/DellEMC_Unity_iSCSI/2470 : SUCCESS in 1h 47m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"ed39064f943ae04f0a65c77d14cefe6998e1f023","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-21 15:37:21.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_PowerStore_iSCSI/dbc22a9/DellEMC_PowerStore_iSCSI/1625 : SUCCESS in 1h 17m 44s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_PowerStore_FC/c881821/DellEMC_PowerStore_FC/1821 : SUCCESS in 1h 44m 20s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_PowerStore_NFS/071f3a0/DellEMC_PowerStore_NFS/1317 : FAILURE in 1h 11m 12s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_PowerStore_NVMeTCP/35f65d2/DellEMC_PowerStore_NVMeTCP/1177 : SUCCESS in 1h 59m 05s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_PowerStore_NVMeFC/a4b63ac/DellEMC_PowerStore_NVMeFC/141 : SUCCESS in 1h 08m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":18},{"id":"6639d8fb171b83763931a70f691c3207f9aced82","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 09:46:27.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_PowerMAX_FC/75345a9/DellEMC_PowerMAX_FC/430 : SUCCESS in 5h 49m 51s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/47/909847/18/check/DellEMC_PowerMAX_iSCSI/85cd717/DellEMC_PowerMAX_iSCSI/452 : SUCCESS in 6h 37m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":18},{"id":"2a57827ca320b88cc21452b88610a5e00e3a1ac0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-23 03:32:52.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/18/20240823083755/NEC-V-ISCSI : SUCCESS in 3h 35m 13s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/18/20240823083755/NEC-V-FC : SUCCESS in 3h 54m 53s","accounts_in_message":[],"_revision_number":18},{"id":"6cf9057a8fb5bcb4399d48505378a1404ca8b701","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-12-01 13:13:47.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/18/20241201162432/NEC-ISCSI : SUCCESS in 5h 21m 18s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/18/20241201162432/NEC-FC : SUCCESS in 5h 49m 12s","accounts_in_message":[],"_revision_number":18},{"id":"586cc46a15009035db66ba53671164f3611082b3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-09 14:23:10.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"7f50700f259421fadda3fd9dd9da8e64a106fdd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 14:42:56.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ea9c7894134726887106e3601c3ad7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73ccf8f2bcec4555a58cd05315e5710f : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/999f86d881854c8f83907881af7ba081 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2820651822bd49fa806d3892456e9932 : SUCCESS in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c70045377fd64d35a451f16f00378c8e47169138","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-09 15:52:29.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/90988f4558fc45d8be21357532ba75fb : SUCCESS in 1h 27m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7dc01d2949184fdbb695dbad2d431f49 : FAILURE in 1h 19m 59s","accounts_in_message":[],"_revision_number":18},{"id":"ed74649d49a7ab9cf2d19ba2ba8fe8e1cf927f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 17:26:22.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0efafe82237a43eb9a2c4feec5822bf9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97a7d1ec33a44c02babb19c47f2e53d0 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/353abf0a0f8641dd895de7419cda46a3 : SUCCESS in 10m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d646d3a06a74cb49ed01c29ded21eb2 : SUCCESS in 13m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d360b67d5444cc69d4cb56793f66f37 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1395d7dc074947b4b906a6cf2cdd6608 : SUCCESS in 16m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/cfda3791baa942db8c4fd93891fecdb0 : SUCCESS in 47m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/053848ca93914a45b90d8f61b7fad642 : SUCCESS in 1h 02m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a159d248e4449acab7d0e599eb5cf84 : SUCCESS in 1h 20m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f98120babcdb451a8bb240445a213fd7 : SUCCESS in 55m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2edc92692c8345859b49d3004d0d62fb : SUCCESS in 25m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a87f1a54b1424e498e589112c9c28caa : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e0621ec32774c028250348e69bd0f6e : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d3ba5c41e31342dcafaa44b008d7edf3 : SUCCESS in 6m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9557661407414a79b26f7fd2046db89d : SUCCESS in 5m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb1806fb859d41a89329c56a714439ab : FAILURE in 30m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfaf54326c7c4798b9c27a5aa2152d4b : SUCCESS in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2146e7642ae4e7fb1e0964d40267a09 : SUCCESS in 1h 50m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b5ff987ecfc4d378a227802f5f7b2f0 : FAILURE in 2h 57m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/677bf1c6ad454537810b22ba0962a479 : SUCCESS in 1h 40m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1c3d4b73e1db457fa9b899adfc7c8a0f : SUCCESS in 1h 56m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c290e0d96f0d4a0b861cacba796a8cda : SUCCESS in 35m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6a55c297b5b84414bc6b2d3e786174f7 : SUCCESS in 1h 03m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/930c99e4e6d14576a2efe0a877671ac6 : SUCCESS in 1h 56m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7b6d39055e4e4bf38453a75c37a905bd : SUCCESS in 35m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/473297de122340bcb63082e0fede56d6 : SUCCESS in 1h 21m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/95d9c972e5a34c90af280d4d5439912c : SUCCESS in 1h 29m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72ce59173aba4b20bf07485db5185939 : SUCCESS in 45m 19s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad149bd7b3544a8c99327953970706c3 : SUCCESS in 1h 32m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cde0acb762c84421a66f3558c3e43dca : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":18},{"id":"ef6e6ab7ae0f5bc48cde4b05795360ab68e5cd4b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-09 17:44:26.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":18},{"id":"8843c0022c1ead1c8057f1b4dacafc8c4351f3c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-24 10:04:16.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"98e1c875ee5883c9265658a66c8dc958128aae53","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-24 10:20:21.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/653/.","accounts_in_message":[],"_revision_number":19},{"id":"1a993303ec4644168555fbc6c73634a4af718719","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-24 10:20:26.000000000","message":"Patch Set 19:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/772/.","accounts_in_message":[],"_revision_number":19},{"id":"1d5d792a602bc5bd19ec439edad215afa62af67e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 10:21:39.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a4ad2ba1a445e3a4e0cf09a0323040\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/63a45f5bd9be425caef99bec1482f50e : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3ef9ac2ed4e41c6999cc9b6275ceeb3 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7013e2d1b0e84444896db220b08609ca : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b984dadc4dbd37ef841853d85a1678ea387b3e96","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-24 10:52:13.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":19},{"id":"829a1fee0cf2d2f43331fcd4f8caf2d57f8bf01f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-24 11:06:03.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":19},{"id":"f21ddc5857d6b49a410d88a85c31d240836a6fae","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-24 11:39:06.000000000","message":"Patch Set 19:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/062898a744bd41c4a397e23631f336bc\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/849bfb1e6a004b52823c65b95949b4a8 : SUCCESS in 57m 50s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4fdba54d1712434bbcf81cfd901a609d : SUCCESS in 43m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7d7196525c9645e3b32885f7ca69ea69 : FAILURE in 1h 34m 09s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b6e9daecd343432ba9bb2ab10855871d : SUCCESS in 51m 24s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8ca102b85b744233a55fef45e81190cc : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6f71d78970b64b63a362532d8b3c1c86 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/044b73a6f7f24a2d8d1fb6a11d44ff5e : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6e8d7f667c34ff3a77f0cfa1bd382a8 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":19},{"id":"695237a1bb6e69734f7502cf50e57306f5704f0e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-24 11:46:43.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/669119372d8c4ff093720939480b0f7f : SUCCESS in 1h 08m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2bfcd1e090b746afbb3731da37b4b5c1 : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":19},{"id":"b4ff8c7eb5a2576fd8a1b8a3e104d3284b756b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 12:47:57.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/217a553113404f63847f5e6b076de4a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d54aedf2214d0892dbf75b8fb8f88e : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ee3a005bb7a466fa58293edd0bf8bfa : SUCCESS in 9m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/accccafd2508491b8a34a2de69f9ecca : SUCCESS in 7m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bf67c25d4e047d69ea7ef4f3ac134ce : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26a4c01f66904c97bf0f02b6d73d0efe : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/bda0f8e2ff0c46868027a372676b7c72 : SUCCESS in 29m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/411b2d4c4e7942d3a1bec83febf5de0f : SUCCESS in 45m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f069c711650a4c9588974e3cdad9d492 : SUCCESS in 1h 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d77e36b665884cc0a96cb387ee9727e4 : SUCCESS in 56m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f984d818cb0468fa95e098eeb176c3d : SUCCESS in 28m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/57211a61ab72405b950cfa9b0de242cf : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17802f089eb84a4dbd045cbd333a2d49 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/70aa8928a9de480dbdebfd6e45a50cbd : SUCCESS in 5m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/312a9cfce52b4e1cbd036acc033794d9 : SUCCESS in 4m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/de15017da23a4e3da9aea48340b68b53 : FAILURE in 27m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f494ee4f1ba2405895017ff676b32054 : SUCCESS in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6b2417084264f4fb5b3f7eb95f49e82 : SUCCESS in 2h 04m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4961972a9efc4398a82012fb680bf001 : SUCCESS in 2h 35m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea7df34c1c6946f993f7ad7faf7c7c3e : SUCCESS in 1h 45m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c1ce0a3408e44ad39e343a3375483221 : SUCCESS in 1h 55m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1f86ccd7532640dabe629d04165067ba : SUCCESS in 26m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55544bbfd7dd4ef2ab1d64e451359abf : SUCCESS in 32m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/953c01e52f4e4780962c0755b0949135 : SUCCESS in 1h 30m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b1a81c87424a4f76ad8e340538608a04 : SUCCESS in 27m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2554cfe27be44b66bf0889660bbf9871 : FAILURE in 45m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1fcd117d8d9b4b29b3f8f99304c53576 : SUCCESS in 1h 52m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35bace3c8bf2441891e4e7f9e86df3fd : SUCCESS in 57m 26s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d5e6a331d594a63a01d5bf0c6a34ade : SUCCESS in 1h 31m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64d679760a1d4cf79246ca04171fafaa : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":19},{"id":"8cd79e94f4bc32bab516fadb9170b62ca6337b14","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-25 21:28:33.000000000","message":"Patch Set 19: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"2a4df88f98397d633a110880195fe721a3ea0b26","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-25 21:29:33.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"4fea538d41b6d67443a32e4c568e30bdfad6b25e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 21:49:47.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d5f6b73a68e4bc8a1e409e2b6c08832\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4b6c38751b245619ed36481b9a68b4f : SUCCESS in 17m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a44cdb0cda5b4bc2a35018f1430e1fae : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0af83eb680c46b1bb4dc3aa0eb10aa7 : SUCCESS in 18m 01s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"eac573f185899af39521c456c8f3747ed278e3e1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-25 22:56:50.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/423d9e4a118c4592b5f0ab92d119f374 : SUCCESS in 1h 12m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/db69911f79a4491eb0cd253a66c32ebd : SUCCESS in 1h 25m 21s","accounts_in_message":[],"_revision_number":19},{"id":"d6affe3a834617cd7731023ffc72d973f9c311c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 23:34:28.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/155aa46c7eac4deca3bd48526259fdb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841fc9bebba64b34b350773910a3a2bb : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d4cade89e4f4fab98ea2ac5de24b31a : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b79cb2afeccb4d68827bc9e93d9b56d1 : SUCCESS in 10m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba9543abc01444d0b71692c51840650f : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69da1c0b995742cbb66310e37715d195 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/2c96499db9a547b0867056aa85704352 : SUCCESS in 52m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ddd91fd90ba4e3bb1ca2ddcb4967899 : SUCCESS in 30m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06eb9c0dd35e425387893dc7c2d99e89 : SUCCESS in 1h 28m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e9e723e34eb4555ba567019c3a43514 : SUCCESS in 35m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7ce30d68a1124a0496ac941f1b5dcbd0 : TIMED_OUT in 40m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2f0c09638bf74b2982b72314065f15c4 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d755a4673154fffaf85d2d312dec51e : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28051da3f2674a0bba9c9b271b352bba : SUCCESS in 6m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a19b0328931e41b484dba02b2e5b5c54 : SUCCESS in 5m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/763492268e284f11bb0cfecb2fd44f07 : FAILURE in 29m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10a798f6ca1d4dae932f5d768d94df94 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/99796cfc952e42da9fd6979002a2dba6 : FAILURE in 1h 30m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/222fc67c799c430f921a30272f120ecb : FAILURE in 2h 03m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3440d32d491948dc98f0cdde23d17827 : SUCCESS in 1h 32m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/50765f363a9e4d18ad4f8bbd0adac52f : SUCCESS in 2h 00m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/97072e1d45884b8f887a6a617ee2304e : SUCCESS in 23m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6813f6619a484e4fb9f2e5ae7b849494 : SUCCESS in 30m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbd600a6fa464e029ccd028baeb3a12a : SUCCESS in 1h 18m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/16ba8609a3d5489d8760a1b46b47eab3 : SUCCESS in 30m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4724013987df4de8b0473a4b2e9f7336 : SUCCESS in 1h 14m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b8bb735f6b864edf8f3263e8997394bb : SUCCESS in 1h 12m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8db9751a0464b448b8f10da916eac80 : SUCCESS in 48m 28s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30576cebd91945dbaf9295e740601c79 : SUCCESS in 1h 37m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72e5ae3918ad45a486608f840873344a : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":19},{"id":"41419a89beb3f7bd2fa9b6a1bc54f68b989915e9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-26 00:39:11.000000000","message":"Patch Set 19:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/19/20250426081808/NEC-V-ISCSI : FAILURE in 46m 50s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/19/20250426081808/NEC-V-FC : FAILURE in 1h 21m 00s","accounts_in_message":[],"_revision_number":19},{"id":"286f755aa70e88788e9e8ea800ee2cc20b83be22","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-27 05:20:00.000000000","message":"Patch Set 19:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/19/20250427125037/NEC-ISCSI : FAILURE in 54m 04s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/47/909847/19/20250427125037/NEC-FC : FAILURE in 1h 29m 20s","accounts_in_message":[],"_revision_number":19},{"id":"cffe442381511689d800ab4f895209e51504e7db","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-28 19:19:09.000000000","message":"Patch Set 19:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":19},{"id":"50f25d8e328210b982cc3f599c57d8fd2e803846","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-28 19:47:18.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"0531f414ff4dd93771f280a377fd17cac6e995a4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-28 19:47:59.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"312272ba3712b990efcfa77cf687b184e7e90955","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:14:47.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c09281ef9b5453f97547b491e594f1a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6152c781f0840849f128a0a815681d6 : SUCCESS in 17m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd7fe00e718a4d33acd664058b73ce98 : SUCCESS in 24m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf9146c33b7d4839a1b67093ed792e9b : SUCCESS in 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"c878271261e7ca0364a6c26307837eced5126cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 22:42:03.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4967144e6a8b4eeebdc9f6fa076feb73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762895b0c6724486811cff692fd7cdf1 : FAILURE in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61be0afeddc540479fe5d222dab77f34 : SUCCESS in 12m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/051e541f1f5f45948b6dc662beb94b4c : SUCCESS in 12m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48a8726b89c344e0b27c0523dcaefbe6 : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfa67d2d1824c7399a127ce071e2fdf : SUCCESS in 6m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/83ae481496a549e0ba736642aeca73a2 : SUCCESS in 49m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/020c478b44ca4acf855264d78636a5b7 : SUCCESS in 56m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a22ee92772434c7cb92629852e9478d2 : SUCCESS in 1h 30m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2180550d438f409b8f851604e73bbf53 : SUCCESS in 1h 05m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bdce5ace3f8c413f926320ab0323e333 : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e33c0aed86b54dfeb0f5c44652bc9801 : SUCCESS in 6m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/57adb5ad988e4f77b33afccd781fc729 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f393fe77fed94ecaa1f4983d295252a5 : FAILURE in 3m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9a6f43a0321248399bf5148341bf7c9e : SUCCESS in 5m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b4f8c3a8377444659c022c78d4bfe289 : FAILURE in 36m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3efd928d356344a69872f17ce75061c4 : SUCCESS in 3m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fe6aacf684fd46c1ade4a036395485b2 : FAILURE in 2h 15m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9cd356ea0ba94802ab8161a951c9828f : FAILURE in 2h 28m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8bf6ac428e744b0797d21b4cf0e45788 : FAILURE in 1h 40m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5b66456e5d5a4009b2d8ed98266c3545 : FAILURE in 18m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1bb623e3e6b04e068dc73269c5262978 : SUCCESS in 32m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/405045fcae714d49954ebc363519d891 : SUCCESS in 59m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e3cc7bda72284072b51f48569b8973c2 : SUCCESS in 1h 30m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/63be7df940d347d89debd678a2eec196 : SUCCESS in 40m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b4ae0473179d4e5aa3e6f352849ec306 : SUCCESS in 1h 11m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/03c5cdffe799421d89c02455dc9c284e : FAILURE in 13m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/432bd47789dd4019a565aac5f449e8d2 : SUCCESS in 1h 49m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fc3eab7ba6f4db3977a27a48def2660 : SUCCESS in 1h 20m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae4ae59f644b494aa88a9d9452168945 : SUCCESS in 56m 35s\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/image/test_glance.py","accounts_in_message":[],"_revision_number":19},{"id":"d899ee408ca0b9badf13773d6c2a7db0cb466edb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-29 00:18:50.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ace91558303c4329a980b2ee575a2164 : SUCCESS in 1h 13m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a7e458067dde43ff83c07b1a6a756e04 : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":19},{"id":"4c400ffc4dc78d5258a1a1d1867d2b80a8d26bac","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-16 07:24:51.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"5fbfc134a495cc2b2eb2ce5ca8ae4e14b75c41c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 07:47:16.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b5605cb6b3a46a091f9dcdb012a2dd4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce38b0d80e3444f09c6885b77f3fb7d7 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d1a9de26724e477eb661b6d6dffad24c : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f965ad06da4a42368a3bdee3ef04ef16 : SUCCESS in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"f6b6201982f1b41cc3a2c9cd9ac178be6be7b576","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-16 08:43:24.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/33ecd6b2997841bcab5b83124b879796 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/df96c08ccd8a485ba20452546d436f8e : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":19},{"id":"d32f769351e478da6063c47b5bcbc1508697d25b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:47:03.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d6b8f61c57d48e58803fe7741c5a101\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3562c2b823ee4ee8a8be2b19d7f064bf : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b11ff75a7464143b2afdab272b403c7 : SUCCESS in 11m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4d3bab266054fd48cc366760f88a459 : SUCCESS in 9m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b39ae4ccfbbc4c9b86131133008e4bd9 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e42e417cef34fbcbee65bedb9e38f21 : SUCCESS in 15m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/69ef60a444354431a4fea02d8dfd3b5f : SUCCESS in 56m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8eb70669dfd2411587bcebef5015c288 : SUCCESS in 56m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd2f62c40dc54f5984623973d85b52df : SUCCESS in 1h 27m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1e6a56358e749fc915202e3c505fffe : SUCCESS in 58m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fdc474c1d78d4a81917da25f1c308f3b : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c320722594ca4fe9856d95f81aba857d : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a5360d7e55ed4627b827c23447faf1ec : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ffcce479cf24a828f2f1a378b220f15 : SUCCESS in 6m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d7e388990fa149cbb6e64ce0a2a68e9b : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/999193ddf20340a0aab508a9aa350c80 : FAILURE in 36m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8494d58d96f840d4947bd7576d7d0fee : SUCCESS in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1774946223e048378168a7884621d137 : SUCCESS in 1h 52m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fb7542ac4b3b456899617e27b2297b1e : FAILURE in 2h 20m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/96dde618a68747558d0656f0313f4f33 : SUCCESS in 1h 52m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/de9f17acfb44480cb11821c9c3f6a5cc : FAILURE in 11m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/70e51ed100954b43885b2bb024628ce8 : SUCCESS in 34m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4a264e02c58249ae88344e9e955268ba : SUCCESS in 1h 10m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/066226185bba4929959fe449c95937b7 : SUCCESS in 1h 32m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/935ce10e1f534eb9bc2f267d202c57ac : SUCCESS in 36m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4e14039e6c294debbf66b6bf08d13e37 : SUCCESS in 1h 46m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/40c77b23f2484122a53fcc0836704c99 : FAILURE in 23m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc483b0ac6a14e5db763b8842f1a8f89 : SUCCESS in 1h 44m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02ac9cc6234e4595859c751ae6165bef : SUCCESS in 1h 21m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ff77d2978d346adbb29ccab53f8cb52 : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":19},{"id":"db1133d08b1058f8b6c7c0453a7055836c47a806","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-05-16 13:01:03.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":19},{"id":"6a27d2747effb729319fb095c44b18c9a8c4a193","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-05-17 23:55:57.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"f78e91b97b3cefd686fbec16dc27945a44da4a6c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-05-18 00:13:13.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"3da5a44afab0a3031b8614ac5a62a9128f8e5dc2","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-05-18 00:29:57.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"6befd8b8b357bba3b328de8e9a8c32d80f394eea","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-05-18 00:46:40.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"b07f9e5a31857eeb69e57abecc2c81be6bf86d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-23 18:36:11.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"9b0521e3cc7a2eda1405e97be5a95deb2b0aa146","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 19:06:02.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497709055f8c4c8da99d7cac1fe3a2c8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e3319f4907b47f1858e7ee8c0dcc549 : SUCCESS in 20m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a44be04800e5438c82b0b0d295053bdf : SUCCESS in 21m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf7355342b814cee961d8d554ac20856 : SUCCESS in 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"fa373f224108ff2b63db93990ade62398a8292ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-23 19:39:46.000000000","message":"Patch Set 20:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3e097570e2d14985a20655288325e337\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/081e325c1d0d478a8c60c9e30259cb13 : NODE_FAILURE Node request 200-0000076294 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d121669d187548d1b41c44ffc63de6e1 : NODE_FAILURE Node request 200-0000076295 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/97c14d453a784ad58af71b25fa19c9f9 : FAILURE in 56m 32s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d61f96cbe5f4049952174675c986733 : NODE_FAILURE Node request 200-0000076297 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4cc810040f0349e0b3257638bcf7667a : RETRY_LIMIT in 26s","accounts_in_message":[],"_revision_number":20},{"id":"6ef853283bd9cc897046f669a587ec3d487c0487","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-23 20:32:40.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f2c4c88ccd724cbab27ae790a03fcffd : FAILURE in 55m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0f658102233d48129f15cffd11e447d5 : SUCCESS in 48m 17s","accounts_in_message":[],"_revision_number":20},{"id":"2ea046342d230c3c636b25f33a94dc153f00f96f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-23 20:46:14.000000000","message":"Patch Set 20:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1307/.","accounts_in_message":[],"_revision_number":20},{"id":"63622d6ed70b72a739759d5da5f8f961f46b60d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 20:50:43.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/746ca8ab1ec1492e99f441fe6ce91f4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19fc1df76484250830b543b4f316d5c : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7946799e8f404e3aa0ad48c663c85eaf : SUCCESS in 11m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d1b2b006b5c459a98e46050a2977b43 : SUCCESS in 8m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb6fddf0d3d9467b9d5c3da932d66b86 : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef08356c62cd45819581162e08dcd7bf : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/54a0ff26949244419743059f95a6b4ff : SUCCESS in 56m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b0c842d745b44c1a3ac46beb3a8e34f : SUCCESS in 1h 01m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07031a0f3ba041a3a19a9f50db24f913 : SUCCESS in 1h 18m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb7f97578b37452e9d4ae5ff0f2b08e1 : SUCCESS in 52m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d3227df31204a5187bc546fccd013c0 : FAILURE in 25m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fa19501bde6b48958e45f491803224cd : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a03503d5eeb4d02a88e538e51a55650 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c58b25e48d7b4513a836d2d173aaf586 : SUCCESS in 5m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5c3c8c18e3084dfe8dc5066a6ea03a84 : SUCCESS in 8m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/819770fdd4bf4774bd090393f4584797 : FAILURE in 25m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c87773a5cc3412eb38ef87bf47b3cf5 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/694a6e6c5c674067b2f78b89aa0412dd : SUCCESS in 1h 33m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/18ad64754afa4df5a518fd658696b1f9 : FAILURE in 2h 11m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cc159fa06d8c404096ee75aaa39af0fe : SUCCESS in 1h 41m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8c7ceb14340c44ee92258198a2939637 : FAILURE in 18m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a5cad910d3984892b25f07f2f26ee344 : SUCCESS in 31m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4705cf00350848918fcde152cdaf68e6 : SUCCESS in 1h 08m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a1ecc128665041e2a99e4f07339d3444 : SUCCESS in 1h 33m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a333d05d50bd4e92bb43fa260ab08905 : SUCCESS in 37m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/42137d0daf3746b8ae95842f56cd5066 : SUCCESS in 1h 29m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/740f1cef91b64a64821ae58ea04e334c : FAILURE in 18m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d7f61b92f00a4b4aa61d8867a502eb91 : SUCCESS in 1h 34m 26s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aae16dcf75de4fddbf2534988e5b91f2 : SUCCESS in 1h 50m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/911fdd9717804de8982f9a25860487b2 : SUCCESS in 59m 12s","accounts_in_message":[],"_revision_number":20},{"id":"2082e30404a698d14a93719d4e1a5825a675ff46","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-23 20:54:20.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909847 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":20},{"id":"3ab8d0a75c8d26506ef78ada794cdbec088d222d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-23 20:56:12.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_07_2025_20_50_54\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":20},{"id":"b39c208fba9c7f4b33361e9b1dfeba40a76880d6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-23 21:08:48.000000000","message":"Patch Set 20:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-47-909847-20/FC/ : FAILURE in 2h 32m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-47-909847-20/iSCSI/ : FAILURE in 1h 51m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":20},{"id":"37a3189a0e199e5446ac3026ff43fd7a8036b8ef","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-23 21:26:30.000000000","message":"Patch Set 20:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1441/.","accounts_in_message":[],"_revision_number":20},{"id":"2ec5ddae052d0950bfca3989f7ffe2092e884040","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-23 21:33:23.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"daf619074a6e0e88b35d75b2315a194eb5f02475","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-23 22:13:50.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909847 : FAILURE in 1h 24m 48s","accounts_in_message":[],"_revision_number":20},{"id":"ca3a17f26cb2ee8af60042fae89135eb3afd3f1e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:17:15.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"f74a94affca31888b596d7e458085f7d7d6ae019","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-25 15:37:16.000000000","message":"Patch Set 20: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"86ee5a2f2c3540254317bcc04ee79424505b1741","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-07-28 14:27:06.000000000","message":"Patch Set 20:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":20},{"id":"fb7bead05bdc948c2f1f830e94293e22416b5b54","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-07-28 19:10:51.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"170f16d36122f41b2c84a67f731de3c565482c74","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-07-29 14:15:10.000000000","message":"Patch Set 20:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"8b1a8df2512e158a326b31eedbd9f66d4a3d9284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 14:15:26.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"c502b587d70d2bdecbd32117d359dc9f7dcc652d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 16:26:59.000000000","message":"Patch Set 20: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909513 is needed.","accounts_in_message":[],"_revision_number":20},{"id":"6b2511a6c6eb2482ae9f6e9a89880ea94a5dc3ac","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-07-29 17:45:27.000000000","message":"Patch Set 20:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/909513 is needed.","accounts_in_message":[],"_revision_number":20},{"id":"4b58e77bfebec57b3a1326770e19bca7c9d71ade","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-29 22:01:45.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2bee858f2bb829bc7b1b0e3d77d7318ba8335a71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 22:30:54.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45f92f4b72642b398b8c603e36bb754\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/728765bae70041ca8302a9af4fedff6b : SUCCESS in 26m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/30a0889170c14c17b6668e2bf1ff62b2 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a53ce7d5b4ec400ebb9c552394170d44 : SUCCESS in 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ba0a28f9a774e035c150e50c42ea564d02b1ad27","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-30 00:05:11.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8aab5ad7b2b54f0f95ac98cf15e7724e : FAILURE in 1h 08m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/034a9d9f4c964e93a28957d1c26bcc00 : FAILURE in 46m 25s","accounts_in_message":[],"_revision_number":20},{"id":"2b974bab0ee1a112f2126f38c115671804ab6ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 00:49:20.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a328e7ea23684a36ba60d2c1d932906d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1993ab6b2f884901bab871402a3442a3 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/369bb2eedf514797a9088ad99e44c5c3 : SUCCESS in 12m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f669ae1c2b5c48b4ac6c5d0752f9bb64 : SUCCESS in 9m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24d109abd92f4ef8987d91260703cc72 : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7bdae854e6347b49f7ac497af92a17f : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/867e6357620d428d934820ae286458e8 : SUCCESS in 43m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7832b984bb1b44a7a66f5f21925375b1 : SUCCESS in 1h 00m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ce0f95f5eb749f9906aeb1e8668f3bb : SUCCESS in 1h 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a19ac239f57c477f87a732514404bc75 : SUCCESS in 1h 00m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0559f86381c2428c8eea83fc96911bf6 : SUCCESS in 23m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e745ce35eeb4a288a3bd7c87899f27f : SUCCESS in 4m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4326eba64cc2447a91253b6fbbf5d557 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28073c5673cb445a91da323af78b75b2 : SUCCESS in 5m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f79c9c9162224ce886e48ee607525b3b : SUCCESS in 10m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/439d02a53c6743348a9cd3d636bb7ad0 : FAILURE in 35m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/007d28ed05984806a775177ecc1b9405 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/06372500916e41418494b14c13a6a1aa : SUCCESS in 1h 57m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aa09092facb747a2a46dfa7d010dfe8f : FAILURE in 2h 45m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c6c8d29fa434998a874639b8335d723 : SUCCESS in 1h 35m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a800cdc413d5463682a0867934701d52 : FAILURE in 21m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e695755bc774314abf78dd91f5991e7 : SUCCESS in 22m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/54f1d0bab61a40ea98cbf13b2af49d6f : SUCCESS in 53m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/268f7b64f97b480b8a8dd57e21b2794e : SUCCESS in 1h 14m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d32734fa7b444e3692c404c9974b2c3d : SUCCESS in 42m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/738d496a189f409cb9fd05dbd2846dbb : SUCCESS in 1h 10m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4867f13a60034711a59eb89ed8c23f4d : FAILURE in 15m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0447f2c4872f4afa9bf064956e1d224e : SUCCESS in 2h 04m 18s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ad7186adcfe4573a26268ca5697c696 : SUCCESS in 1h 15m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e108d2125b53486cb0da84ce1264cb48 : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":20},{"id":"fdd9fb0c2398cd573c49a82c6ab71ea77d8c2950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 00:49:57.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"48b66be8d3a0c3701fbc650a5e98ca95e03e86e3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-30 00:50:02.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"5b561d77cc4ebd6f9456f842c406b1071e612ec0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-30 00:55:09.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_07_2025_00_49_31\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":20},{"id":"534b7606e40abf75619e541109bafc10db8f0ae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 02:43:47.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cae5ed0378145c4be60d7ae5581e440\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7621c55db86c48418c67de34802633a6 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21bfe14eb7194c07bb6956c2155c4eef : SUCCESS in 6m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dbcbb24e7f4949eeb3df451f99716e9b : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2938b1946d0841a1ae17fc3e10318f9c : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c8e22e1544844c709ba7e89c78a176a7 : SUCCESS in 29m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efe70e75ee8642b8b658d7888aeb6e8c : SUCCESS in 31m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4c89a41c2544b9fb489631e79309aa5 : SUCCESS in 1h 23m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68bd9436375a45789f341de386ddffed : SUCCESS in 1h 01m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/02e53ddf61a04acebf27177864d6dc68 : SUCCESS in 1h 14m 11s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/26299854e3694e9e92bd09bbd77c8216 : SUCCESS in 1h 51m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a705bca8e63f40b7b3733b59ad35746a : SUCCESS in 1h 23m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e1fa8138f6744fca898c04e0a62567e : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":20},{"id":"fd9baf9aed0a2df5b264a143c9ec13b23a8ff230","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 02:43:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"7b276aab6bb759b2aa71dea4afdfad19a954d777","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 02:46:06.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3fddbdbbc924f468c998b5798d196bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8af317d6fd0430d928e7f4d5533a9ca : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"787ca3db096fca034fb070fc77a2c55630bcea80","revisions":{"2b7b42dc6c6d184f574e7de12ea0e5b533d4c19f":{"kind":"REWORK","_number":1,"created":"2024-02-22 09:12:43.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/1"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:12:24.000000000","tz":330},"subject":"Add tests for optimized volume upload","message":"Add tests for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b7b42dc6c6d184f574e7de12ea0e5b533d4c19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b7b42dc6c6d184f574e7de12ea0e5b533d4c19f"}]},"branch":"refs/heads/master"},"bbe558ece661e584e8791b3020690939b9680016":{"kind":"REWORK","_number":2,"created":"2024-02-22 12:58:18.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/2"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:32:56.000000000","tz":330},"subject":"Add tests for optimized volume upload","message":"Add tests for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbe558ece661e584e8791b3020690939b9680016"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbe558ece661e584e8791b3020690939b9680016"}]},"branch":"refs/heads/master"},"292a1cf64e54c6c921d8b7eda36b9f40ac6bb1ff":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-02-26 09:15:32.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/3"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-26 09:15:32.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\nDepends-On: https://review.opendev.org/c/openstack/glance/+/906359\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/292a1cf64e54c6c921d8b7eda36b9f40ac6bb1ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/292a1cf64e54c6c921d8b7eda36b9f40ac6bb1ff"}]},"branch":"refs/heads/master","description":"Edit commit message"},"15a5ad68170af14d3df1bf37ce204030217edc64":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-02-27 06:01:09.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/4"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-27 06:01:08.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\nDepends-On: https://review.opendev.org/c/openstack/glance/+/890509\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/890634\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/15a5ad68170af14d3df1bf37ce204030217edc64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/15a5ad68170af14d3df1bf37ce204030217edc64"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4c1371c91d62bab5daefb0d98312a1674748629a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-02-27 07:26:30.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/5"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-27 07:26:30.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/890634\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c1371c91d62bab5daefb0d98312a1674748629a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c1371c91d62bab5daefb0d98312a1674748629a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d68e090261509a569588a2917619a23944348b1f":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-02-27 07:28:49.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/6"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-27 07:28:49.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/glance/+/906359\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/890634\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d68e090261509a569588a2917619a23944348b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d68e090261509a569588a2917619a23944348b1f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3f40b9575dbdb18c5edc7c06cabc94cb35f18641":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-02-28 03:57:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/909847/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/7"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 03:57:00.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/glance/+/906359\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f40b9575dbdb18c5edc7c06cabc94cb35f18641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f40b9575dbdb18c5edc7c06cabc94cb35f18641"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eff4b66ea00f489edb02d4d6c34bc5aaeb549e52":{"kind":"REWORK","_number":8,"created":"2024-02-28 04:32:42.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/8"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-28 04:32:42.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/glance/+/906359\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eff4b66ea00f489edb02d4d6c34bc5aaeb549e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eff4b66ea00f489edb02d4d6c34bc5aaeb549e52"}]},"branch":"refs/heads/master"},"c2e94709da27ccd5c71d49efb2e118552be6634b":{"kind":"REWORK","_number":9,"created":"2024-02-28 05:35:21.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/9"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-28 05:35:20.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/glance/+/906359\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2e94709da27ccd5c71d49efb2e118552be6634b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2e94709da27ccd5c71d49efb2e118552be6634b"}]},"branch":"refs/heads/master"},"c5525bc7534e008b162cdd0bb6c969602d0e8ace":{"kind":"REWORK","_number":10,"created":"2024-02-28 07:44:12.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/10"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-28 07:44:11.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/909841\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/glance/+/906359\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/910473\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5525bc7534e008b162cdd0bb6c969602d0e8ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5525bc7534e008b162cdd0bb6c969602d0e8ace"}]},"branch":"refs/heads/master"},"44243ed4af487917ac04471cf34164126b18ae16":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2024-02-28 07:57:50.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/11"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-28 07:57:50.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/910473\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/glance/+/906359\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44243ed4af487917ac04471cf34164126b18ae16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44243ed4af487917ac04471cf34164126b18ae16"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c7163d3283892dbf7cf6e3fb101a518d0afb8869":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2024-02-28 19:09:13.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/12"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-28 19:09:12.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/910473\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7163d3283892dbf7cf6e3fb101a518d0afb8869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7163d3283892dbf7cf6e3fb101a518d0afb8869"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7c47643dab6ec332986763e4951eac2c12a724fd":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2024-03-04 22:08:58.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/13"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-04 22:08:58.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/910473\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/911004\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/911005\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c47643dab6ec332986763e4951eac2c12a724fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c47643dab6ec332986763e4951eac2c12a724fd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b1b937b3cf45e4eea3cb5649a1aff9733bc16d8b":{"kind":"REWORK","_number":14,"created":"2024-03-05 07:10:32.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/14"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-05 07:10:31.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/910473\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/911004\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/911005\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1b937b3cf45e4eea3cb5649a1aff9733bc16d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1b937b3cf45e4eea3cb5649a1aff9733bc16d8b"}]},"branch":"refs/heads/master"},"d52f8a59c922f3398eca40271ee42997f679265d":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2024-03-05 08:35:51.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/15"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-03-05 08:35:51.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/910473\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/911004\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/911005\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/911047\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d52f8a59c922f3398eca40271ee42997f679265d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d52f8a59c922f3398eca40271ee42997f679265d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"036b53e0a9574534e5864033d2aee04076533e05":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2024-07-15 20:27:13.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/16"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-15 20:27:12.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/910473\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/911004\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/036b53e0a9574534e5864033d2aee04076533e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/036b53e0a9574534e5864033d2aee04076533e05"}]},"branch":"refs/heads/master","description":"Edit commit message"},"88f00c6a77a4834f7aa981d53c4d2358ec9095eb":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2024-08-19 22:58:39.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/17"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-08-19 22:58:39.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\nDepends-On: https://review.opendev.org/c/openstack/python-glanceclient/+/899453\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/911004\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88f00c6a77a4834f7aa981d53c4d2358ec9095eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88f00c6a77a4834f7aa981d53c4d2358ec9095eb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"49ed2f799ae65cfc070957bca87dcda8cd1b608f":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2024-08-21 08:31:11.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/18"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-08-21 08:31:11.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49ed2f799ae65cfc070957bca87dcda8cd1b608f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49ed2f799ae65cfc070957bca87dcda8cd1b608f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d56cee1e35d3ba3f8e8f31c2b394cab28e64fd42":{"kind":"REWORK","_number":19,"created":"2025-04-24 10:04:16.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/19"}}},"commit":{"parents":[{"commit":"0425c08f3d19ee0487c26bec499cf84e4538f519","subject":"Merge \"Hacking: Remove C306, C308 checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0425c08f3d19ee0487c26bec499cf84e4538f519"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-04-24 10:04:15.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\n\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d56cee1e35d3ba3f8e8f31c2b394cab28e64fd42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d56cee1e35d3ba3f8e8f31c2b394cab28e64fd42"}]},"branch":"refs/heads/master"},"787ca3db096fca034fb070fc77a2c55630bcea80":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":20,"created":"2025-07-23 18:36:11.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/909847/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/909847/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/909847/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/909847/20"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-22 09:08:30.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-07-23 18:35:10.000000000","tz":0},"subject":"Add testing for optimized volume upload","message":"Add testing for optimized volume upload\n\nWhen using cinder as glance backend, we have an optimization for\nupload volume to image operation.\nIt was recently found that the glance location API was broken for\nthe upload volume optimization. The current glance cinder job\ndoesn\u0027t run the upload volume test hence the issue was not detected.\n\nThis patch adds the volume action tests (that includes the upload\nvolume test) in the glance cinder job. It also requires configuring\nsome parameters that are set up in devstack in the depends-on patch.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/909513\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: I4918027ed641a90aafe44d815a4a3dbc1dc7ddfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/787ca3db096fca034fb070fc77a2c55630bcea80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/787ca3db096fca034fb070fc77a2c55630bcea80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
