)]}'
{"id":"openstack%2Fcinder~910234","triplet_id":"openstack%2Fcinder~master~If3e94ea2b192c9d7fef2e96325563819c8c15935","project":"openstack/cinder","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-08-20 12:12:42.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2025-08-20 12:12:42.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-02-29 14:08:53.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If3e94ea2b192c9d7fef2e96325563819c8c15935","subject":"mypy: Add minimal sqlalchemy/api.py coverage","status":"NEW","created":"2024-02-26 14:39:04.000000000","updated":"2025-08-20 17:27:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc4ed8db6cf69726d8ca6fa87ea386a5bd99dca0","_number":910234,"virtual_id_number":910234,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-20 15:19:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 14:39:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 14:50:53.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-26 14:59:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 15:16:29.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 15:20:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-26 16:08:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-26 16:22:27.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 16:30:53.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 17:17:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 17:59:48.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-26 18:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-26 18:56:49.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-26 23:19:39.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-27 03:46:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-27 04:36:59.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-27 07:09:55.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-27 08:09:09.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-02-29 17:12:36.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-29 18:06:01.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-18 04:07:58.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2025-08-20 12:12:51.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-20 12:13:35.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"}],"messages":[{"id":"35f08f92b114a7e1280b2051990859d9205101e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-26 14:39:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d0c0a0312cdf3c8bb92a75412beb54ecd059307","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-26 14:39:16.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16556/","accounts_in_message":[],"_revision_number":1},{"id":"31ab7d38c27f6e35ccc9946ba4e2ae2b34a7b36c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-26 14:39:22.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16556/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"50f7456fcba408c802494ac49028ae1619a955ba","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-26 14:50:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4810a09a5759484c93c381558ee123cdf0ccb91e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 14:59:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90081098ac2643a192467449eccfb727\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc192232273f4847bd88ad496be98b7b : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6a346dc620b6466bb05ba29a9ead1465 : SUCCESS in 15m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9c3dde2e15e4c349878bbd8b4522678 : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1f4f5d674a95f59210e7359d0b25769437b826e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-26 15:16:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910234 : FAILURE in 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e32b38a73ca9b248149cb744481ed7a54b05940f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-26 15:20:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910234 : FAILURE in 40m 00s","accounts_in_message":[],"_revision_number":1},{"id":"62648c18f444f9b791dc627019d3a94b1bca427c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-26 16:08:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/785/.","accounts_in_message":[],"_revision_number":1},{"id":"d426db853359b054ec84295ba5e7ca6c14e3f15a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-26 16:22:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/34/910234/1/ : SUCCESS in 1h 34m 49s","accounts_in_message":[],"_revision_number":1},{"id":"77decc43839898df927a9118ea4477f33019e16f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-26 16:30:53.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/34/910234/1/check/DellEMC_PowerFlex_3.6/4289204/DellEMC_PowerFlex_3.6/1531 : SUCCESS in 1h 31m 00s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/34/910234/1/check/DellEMC_PowerFlex_4.5/94cd698/DellEMC_PowerFlex_4.5/495 : SUCCESS in 1h 51m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"cbdcfb3f19594da5b073c019447ac1998faf7686","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-26 16:49:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910234 : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":1},{"id":"45159cb1f9df73a333b1510297e7429e3dcbdb64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 17:17:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b10a7edba0704b058b0752cbb388cf51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/569b082e8b224877b8afc416b7e04e74 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1edbcdac1e74561961d225a919ea424 : SUCCESS in 10m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70ac89e3ae334e0fb5674ae83b6ff4e6 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f51ca4d0a745c385b2131ff40519bd : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/6012d43a82154e9b9fc95345de910ba6 : SUCCESS in 1h 08m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7fab3c4abf2c4179892f94002bf8df5b : SUCCESS in 1h 01m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe41b4ff4a0a4569b9f9e3d61590bb1c : SUCCESS in 1h 20m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be41b240eedc4a1e8e8f3370fb350843 : SUCCESS in 1h 08m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0a03abb50dda4e3285551a8ad3c505bf : SUCCESS in 17m 27s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/26891dc0cbdd42a884ba28f244cf6427 : SUCCESS in 13m 53s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7124226f1a5d4574a45e9984a5a19ed2 : FAILURE in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/569b87be3f9c40eb9852199f3a8d88fe : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/86a0dba69cdc4ef8ade540f7e1426a01 : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d5dc37b9d4547b6bbec13ad98ad7dbb : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fd76b42dce304580b321b24fcc44dd41 : SUCCESS in 37m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50dac698f973490c9941ed8e369929aa : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/790f9fd40f034217a244c77c927d25a6 : SUCCESS in 2h 10m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2123e0f8e0a84d49a638722bbd5745dc : FAILURE in 2h 32m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e40f9e8063c84682aec5d84be23ce958 : FAILURE in 1h 59m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/784a293e438749ce969b5921ccff8931 : SUCCESS in 1h 41m 01s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c87552be51144b8a90aaa70cbf25335 : SUCCESS in 1h 12m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58e73950585e4e01a53d8486c20f2847 : SUCCESS in 1h 33m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/64074a7b3b0845c1ab89d22f34babe89 : SUCCESS in 43m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25ad951c547741b7a235dc30d862c8e1 : SUCCESS in 1h 26m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c4f7677f89d347f996686d39491e56ae : SUCCESS in 1h 40m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a7f98b580094feda8ec78a4184948f0 : SUCCESS in 1h 28m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7464dfaa605409d91cbc1ace6c1a476 : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5099e1b88cd5b296ea614c5a714988e44cee1f10","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-26 17:59:48.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9850444b7023411fb9120ee5bb57faec\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/359f59bdf28e4b6cb2ba48c6d0d641de : RETRY_LIMIT in 1h 30m 45s","accounts_in_message":[],"_revision_number":1},{"id":"55e48e241af99d7883fc9d77660ca4cd9d309b82","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-26 18:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/910234/1/fc : SUCCESS in 0h 54m 57s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/910234/1/iscsi : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":1},{"id":"72f0b67682335dcade6baa16bd7de11575c5ad9b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-26 18:56:49.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-910234-1/FC/ : SUCCESS in 2h 26m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-910234-1/iSCSI/ : FAILURE in 4h 17m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"d474d6570c7380f7fac089f2c5bfd4cd9a146e4c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-26 23:19:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/910234/1/20240227022431/NEC-V-ISCSI : SUCCESS in 2h 55m 18s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/910234/1/20240227022431/NEC-V-FC : SUCCESS in 5h 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"61294238968166cd62407bbf268e17744445d902","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-27 03:46:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/910234/1/check/DellEMC_Unity_FC/d84c5e5/DellEMC_Unity_FC/1392 : SUCCESS in 1h 14m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/910234/1/check/DellEMC_Unity_iSCSI/3185762/DellEMC_Unity_iSCSI/1425 : SUCCESS in 1h 35m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c436a54ab05584e8aa0543ee49c5c039bc7a29b9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-27 04:36:59.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/910234/1/20240227022440/NEC-FC : SUCCESS in 10h 52m 25s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/910234/1/20240227022440/NEC-ISCSI : FAILURE in 11h 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5d5a0f75b782404fb2abec3db993b0a9931a5e4a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-27 07:09:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bf9cfef6535f421fb0661681161f4e51 : FAILURE in 1h 11m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9a2a3b21476d4a339b1382f7e6c9c9a5 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":1},{"id":"dea9b47b6ea3106ab536911993134be9932467a0","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-29 13:08:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4efc505118fac56902f3c61b747f3027ac2f8e84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-29 13:08:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"11f8cf2ea88ee713bd6c283a4a960cbc5890056f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:32:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0136e79e8c2248bcaf5291a0c24b21bd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fec7f47812e4b44a528d53ab5cdf927 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78ff91412d5942778b2bcb59a44c8493 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/08708a4381e3414dbdd5d216d2bc1ae9 : FAILURE in 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9bcebfa85a19d2794fc8b03dce6e1e7d3aedfd37","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-29 13:59:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/34/910234/2/ : FAILURE in 42m 28s","accounts_in_message":[],"_revision_number":2},{"id":"0fc28b9617691ac266a59f7b2f8e1a02f0411dbd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-29 14:05:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910234 : FAILURE in 55m 05s","accounts_in_message":[],"_revision_number":2},{"id":"272dbb9855f62d7b975a0e826bc1cfbc02891e62","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-29 14:08:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910234 : FAILURE in 57m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7370d5b9a325d970c8c85d8dd71e39dfa49c65b1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-29 14:21:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ed4bb378347040e78f8bc27df31ff791 : SUCCESS in 1h 11m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e6946294bd4e42da8a7ca62a81c48f7f : FAILURE in 1h 07m 55s","accounts_in_message":[],"_revision_number":2},{"id":"9bf0e72c044247483e3cc40a17c7f4a1d6541e00","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-29 14:40:32.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_Unity_FC/6633b1a/DellEMC_Unity_FC/1434 : SUCCESS in 1h 30m 05s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_Unity_iSCSI/d89bda7/DellEMC_Unity_iSCSI/1466 : SUCCESS in 1h 31m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"aa86cd65bec675c5571126daf4dcebc63345b5a0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-29 14:40:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_PowerFlex_3.6/3bf41b7/DellEMC_PowerFlex_3.6/1577 : SUCCESS in 1h 32m 03s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_PowerFlex_4.5/df568d8/DellEMC_PowerFlex_4.5/530 : FAILURE in 48m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"81422e7e07d3fe8abce8b857e944a6243db65e86","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-29 15:30:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910234 : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6b59c4b51cfc0519864f423e95f3987e2e9e6221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 16:08:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e166bd68ce400e896ffbe3d9d8cef6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00be239d7ccf4863842340f8379295ca : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1a67c66a556488cb40c0abbaf1beb40 : SUCCESS in 11m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/457e9b4aa19a480b9ab4935562b9c6ee : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a5d4245e4549e990c1b6959a56c8b6 : SUCCESS in 11m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/dc21f479026b44bda370f4913978220f : SUCCESS in 56m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/297dfca7a17e4866bb51ede94036f156 : SUCCESS in 55m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2e0dcb44af1e48bbaf8b69ca84f175f1 : SUCCESS in 1h 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49db25adb4b14b3b9ef2dcad8ec027ac : SUCCESS in 57m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b79bcbca9758479ca5f7a598f7fbe2b1 : SUCCESS in 13m 58s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/27f3dfbe97294cf38a440fa5e16f0317 : SUCCESS in 12m 00s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5d82031350e4890a4affdcf5accff34 : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2ae5178751354688b2ea3028b88b6b9e : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb3bd4b79f8946b1bbd4452155d17146 : SUCCESS in 6m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1a4d9378f3224df8bc87846f25b7d314 : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5fd5777e61f74a8eb9d70f52238be40d : SUCCESS in 45m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a4c96b140e84c82a1dea86347f5c445 : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dbd5e62e72db4eab8bc364dd708de206 : SUCCESS in 2h 05m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/152a1bc1d78943e08cb8ef24be19ee8c : SUCCESS in 2h 24m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3e23babd84be460988bc3bafcb3b3df3 : SUCCESS in 1h 30m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3d4c98f54fe4459ca8aaeba2755cddfe : SUCCESS in 2h 16m 42s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3661b836e39e44ed82f710d54abd7d6d : SUCCESS in 1h 14m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6846a04272754de78250cae31daef3f5 : FAILURE in 1h 50m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f940ec0d69f1437d8b76abd7aac58b30 : SUCCESS in 47m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cd0e383eef464398a090cc15ac14a08f : SUCCESS in 1h 30m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/025707aa43a0451a8047f6a04491b02a : SUCCESS in 1h 43m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d678108db3234e70bc7d759ed562611f : SUCCESS in 2h 05m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/065d18df571d4bd0bd6ef57b6d196956 : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5c521ab8addb93722527213abeb898c549d650c8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-29 17:12:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/910234/2/upstream-check/cinder-tempest-plugin-ontap-fcp/830e15a/ : POST_FAILURE in 57m 55s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/910234/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/8abcd96/ : FAILURE in 11m 02s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/910234/2/upstream-check/cinder-tempest-plugin-ontap-nfs/c28f5e4/ : FAILURE in 11m 56s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/910234/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8c7e9a0/ : SUCCESS in 40m 10s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4d72d742a8048e624722dce69a260819449c30c0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-29 17:49:45.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_Unity_FC/88a867d/DellEMC_Unity_FC/1436 : SUCCESS in 1h 11m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_Unity_iSCSI/ec4ea04/DellEMC_Unity_iSCSI/1468 : SUCCESS in 1h 37m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"3a12e099a19267ce65f42dcd9c2c4c6fbcbfa8f5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-29 18:06:01.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_PowerStore_iSCSI/35a50f6/DellEMC_PowerStore_iSCSI/1132 : SUCCESS in 1h 06m 34s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_PowerStore_FC/7f59b0d/DellEMC_PowerStore_FC/1297 : SUCCESS in 53m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_PowerStore_NFS/506c4ba/DellEMC_PowerStore_NFS/816 : SUCCESS in 1h 05m 41s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_PowerStore_NVMeTCP/6452aee/DellEMC_PowerStore_NVMeTCP/684 : SUCCESS in 1h 57m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"0f89f54f53315be780ff0219b292b484d7fce47b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-29 18:24:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_PowerFlex_3.6/829d842/DellEMC_PowerFlex_3.6/1579 : SUCCESS in 1h 43m 06s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/34/910234/2/check/DellEMC_PowerFlex_4.5/9f61c8a/DellEMC_PowerFlex_4.5/532 : SUCCESS in 2h 15m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"0a8ac74d57e3dbaba414584b3330421bed961e8a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-29 20:13:20.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-910234-2/FC/ : FAILURE in 6h 37m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-910234-2/iSCSI/ : FAILURE in 7h 4m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"004009779c787fe3c049fc6ce7c76b4c33a74dc3","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-01 04:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/910234/2/fc : SUCCESS in 0h 50m 10s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/910234/2/iscsi : SUCCESS in 0h 45m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4f26d249b91249d914a81c362c80e57066a228d2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-02 18:43:04.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/0d0a3a6dfcd044dcb5fb4a88b024e491\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f90fdc27a7ab4cf480638bd2f0e02827 : RETRY_LIMIT in 24m 06s","accounts_in_message":[],"_revision_number":2},{"id":"901300d5d40118767d06d16552f3b10c6eafa517","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-02 23:31:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/910234/2/20240303001522/NEC-ISCSI : SUCCESS in 4h 00m 19s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/910234/2/20240303001522/NEC-FC : SUCCESS in 8h 15m 55s","accounts_in_message":[],"_revision_number":2},{"id":"a174b9bb31725c7636522c038fa6dced0fc9770a","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-05-18 04:07:58.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce1f1e8144744252704a2a09dc7632ddc8688db7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-22 13:35:51.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/910234/2/20240722150108/NEC-V-FC : SUCCESS in 3h 52m 49s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/910234/2/20240722150108/NEC-V-ISCSI : FAILURE in 7h 34m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d3bb29a6bffcc5abaf28142c8389dc52215b6ed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-08-20 12:12:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b3338bfe87ed30cf296a9a0230fa8bd0fcf22e84","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-20 12:12:51.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/223/","accounts_in_message":[],"_revision_number":3},{"id":"0f9317676882ea623d17e912622281933b9c3222","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-20 12:13:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"6c6bcbe4a6ec4efdd50796844be11e7d35ee31f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-20 12:27:25.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910234/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910234/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"c06bfa4c1b21eef062066d7d737218b5fc1311dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-20 13:06:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1627/.","accounts_in_message":[],"_revision_number":3},{"id":"300402184341aa2301f7620218cc1609500cef32","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-20 13:07:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1492/.","accounts_in_message":[],"_revision_number":3},{"id":"389bd0c1a00e47f3f583f0d25d6aaa6302e44e70","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-20 13:39:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910234 : FAILURE in 1h 25m 20s","accounts_in_message":[],"_revision_number":3},{"id":"8f36c9cd9a530e86d6e069b364653bba91a36415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:52:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ab0180c5714e7999e93eafef09f11a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/460e0eb797ee46aa9fd91bdbd94f839b : SUCCESS in 17m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd2ce78431934a7d82be0fe10fc621e2 : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7850597a3b324d8a81228ef3668e5275 : SUCCESS in 42m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7350532edf2ef0ac5f133dfaff5b520cbe457138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 15:19:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de143a9e8f0547128b8901537bdb4eca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca80fbeb22944248b2321f7f8386a8a : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f589d2aec484a9ba907ea3416761817 : SUCCESS in 12m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02c0dd71f5e84a599251298a4679f818 : SUCCESS in 13m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfd089595d534ce1a21d1e2d3eaded7d : SUCCESS in 17m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f490136f6ddb430cae13d5d59de4eb22 : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/336e086250e64aeea9a57f6741316207 : SUCCESS in 49m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d6cb3d23b2b4fdfb83ad7cfa222deb2 : SUCCESS in 54m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ebcc171267f4ddf96a92b9c8fa79e41 : SUCCESS in 1h 40m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0634ca5102034a9fa7f6563512d0a2a7 : SUCCESS in 34m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/023f85cd139b40609eebe6613d55102b : SUCCESS in 16m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9967603c6789455abebef4d26f9d8bce : SUCCESS in 3m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9df0d7ad58749f1a3d1d62693962efe : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/021139d462584e348a79db991be97260 : SUCCESS in 5m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fdf95d8b82e845048ec244607daa3631 : SUCCESS in 9m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0c6f9df6fff747dcb736b922b65e8f76 : FAILURE in 19m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/707db51f11cf4050a9fc23107cb6149b : SUCCESS in 3m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5db9a2003ab14fc6a96cfc669a02cae0 : SUCCESS in 2h 02m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/22f8036bfa3b4e7a9b981d89ea4f3eee : FAILURE in 3h 00m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ebb655d2641f481a860c2e27fbf6ddb5 : SUCCESS in 1h 26m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0fb411f6a72b47f5b35ec48aacf3dc1c : FAILURE in 23m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f99ab25ef52547b58c7a972a9c26e98f : SUCCESS in 40m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b93ca93ac7c402791b13632de660e8a : SUCCESS in 1h 18m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e7fd692398214483b319fe98ebffcb2a : SUCCESS in 1h 58m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d7de3576a2cb42f89c3089f1a6324e06 : SUCCESS in 54m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c6892f26a3dd446fa6ce5bab2507ecd0 : SUCCESS in 1h 42m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/de0fb874420e47df9ae2c6ae9880762a : FAILURE in 11m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20343686b4df48598b5cb2623b8a3add : SUCCESS in 1h 30m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47bf88cb2e0e42e58f34822c22f0987a : SUCCESS in 1h 13m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a424cfbbe1b4df7ad8b482a56d0a854 : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":3},{"id":"7ec12f59a5e45f3e7698a5bf1693d67788282e78","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-20 15:54:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d7b8127deedb4b0db5664d2b674ad808 : SUCCESS in 1h 36m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/031345aea78a49d5855bfa43f9ec6a03 : FAILURE in 2h 07m 55s","accounts_in_message":[],"_revision_number":3},{"id":"dc4ed8db6cf69726d8ca6fa87ea386a5bd99dca0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-20 17:27:34.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-910234-3/FC/ : FAILURE in 5h 11m 34s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-910234-3/iSCSI/ : FAILURE in 3h 46m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5a93be03f87c60edbe0da2f296ae25a26056c795","revisions":{"b4cf0249f2bd66532746b53726a6ad97bbb71376":{"kind":"REWORK","_number":1,"created":"2024-02-26 14:39:04.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/34/910234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/910234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/910234/1"}}},"commit":{"parents":[{"commit":"647fa0b10222c919dbeeeb19b761b5521fd01961","subject":"Merge \"Add the os-extend_volume_completion volume action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/647fa0b10222c919dbeeeb19b761b5521fd01961"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-26 14:37:04.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-26 14:38:09.000000000","tz":-300},"subject":"mypy: Add minimal sqlalchemy/api.py coverage","message":"mypy: Add minimal sqlalchemy/api.py coverage\n\nThis catches the bug reported as bug #2047693.\n\nRelated-Bug: #2047693\nChange-Id: If3e94ea2b192c9d7fef2e96325563819c8c15935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4cf0249f2bd66532746b53726a6ad97bbb71376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4cf0249f2bd66532746b53726a6ad97bbb71376"}]},"branch":"refs/heads/master"},"98002806e86e5abfdfc521aa6404c5a06c4d7726":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-29 13:08:33.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/34/910234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/910234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/910234/2"}}},"commit":{"parents":[{"commit":"ba424aae9241551433f80c15555f815ba9c9d8ac","subject":"Merge \"Dell PowerFlex: Added timeout into rest API call.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba424aae9241551433f80c15555f815ba9c9d8ac"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-26 14:37:04.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-29 13:08:33.000000000","tz":0},"subject":"mypy: Add minimal sqlalchemy/api.py coverage","message":"mypy: Add minimal sqlalchemy/api.py coverage\n\nThis catches the bug reported as bug #2047693.\n\nRelated-Bug: #2047693\nChange-Id: If3e94ea2b192c9d7fef2e96325563819c8c15935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98002806e86e5abfdfc521aa6404c5a06c4d7726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98002806e86e5abfdfc521aa6404c5a06c4d7726"}]},"branch":"refs/heads/master","description":"Rebase"},"5a93be03f87c60edbe0da2f296ae25a26056c795":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-08-20 12:12:42.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/34/910234/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/910234/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/910234/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/910234/3"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-26 14:37:04.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-08-20 12:12:36.000000000","tz":-240},"subject":"mypy: Add minimal sqlalchemy/api.py coverage","message":"mypy: Add minimal sqlalchemy/api.py coverage\n\nThis catches the bug reported as bug #2047693.\n\nRelated-Bug: #2047693\nChange-Id: If3e94ea2b192c9d7fef2e96325563819c8c15935\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a93be03f87c60edbe0da2f296ae25a26056c795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a93be03f87c60edbe0da2f296ae25a26056c795"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
