)]}'
{"id":"openstack%2Fcinder~910320","triplet_id":"openstack%2Fcinder~master~Iafa44dcf0e8f46749b5ef37f0b8d341e8253a3bd","project":"openstack/cinder","branch":"master","topic":"bp-powermax-protected-snap-config","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-03-02 15:00:22.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2024-03-15 07:39:29.000000000","reason":"\u003cGERRIT_ACCOUNT_31779\u003e replied on the change","reason_account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}},"17450":{"account":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"last_update":"2024-03-15 15:07:32.000000000","reason":"Change was submitted"},"36396":{"account":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"last_update":"2024-03-15 15:07:32.000000000","reason":"Change was submitted"},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2024-03-15 15:07:32.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-03-11 18:18:16.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Iafa44dcf0e8f46749b5ef37f0b8d341e8253a3bd","subject":"Dell EMC: PowerMax - Configurable SRDF snapshots","status":"MERGED","created":"2024-02-29 00:45:28.000000000","updated":"2024-03-28 12:41:56.000000000","submitted":"2024-03-15 15:07:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":59,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910320-bp-powermax-protected-snap-config","meta_rev_id":"d221db220ae232f5a651fb65d12c05774b92b812","_number":910320,"virtual_id_number":910320,"owner":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-15 15:07:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-03-11 18:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-11 18:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-03-11 18:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 01:15:38.000000000","updated_by":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"reviewer":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"state":"CC"},{"updated":"2024-02-29 01:15:38.000000000","updated_by":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-02-29 01:15:38.000000000","updated_by":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"reviewer":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"state":"CC"},{"updated":"2024-02-29 01:15:38.000000000","updated_by":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2024-02-29 01:15:38.000000000","updated_by":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2024-02-29 01:15:38.000000000","updated_by":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"CC"},{"updated":"2024-02-29 01:15:38.000000000","updated_by":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-02-29 01:15:38.000000000","updated_by":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-29 01:24:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 02:25:51.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 02:27:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-29 02:54:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-29 02:55:46.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-02-29 03:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-29 04:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-29 04:03:23.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 04:24:14.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 06:19:54.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-29 14:34:03.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 15:33:19.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 15:33:43.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 16:48:08.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-02 18:25:58.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-02 19:40:12.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-03 21:26:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-11 14:42:24.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-28 12:41:56.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"894315764807cd7c7943e2915dd8ff5e5b5dd2e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 00:45:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb7a620efbfc044af51ccafde84dc7a9faa0e812","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 00:47:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9101762b824b7e9df1fe3c30a3fec6c0e12dd6f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 00:52:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1632a8d9b3c1162b85c953c1d362e02fcc6495d2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 00:53:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"79311c4ba190cc2c58aa28af3278a25359f7c756","tag":"autogenerated:gerrit:restore","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 00:55:24.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"2e5b0d2c54ca18ffcd9e8bcef627318d5f6df77a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 01:12:42.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"ffbd75192ead85129ebf2ced93a026d38227b011","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 01:24:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0b03693f9c54a729b155bf6e1b69dd8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcaf58bf9c024c09aa9d6c2a20d49231 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4bb54d7a92745a3b3055931a04129de : FAILURE in 29m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a5d758286824f58a65ab30cff616db7 : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aaed6b2b6871a9c2ad298242efc4901a94d1ed0e","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 01:49:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e243bb058a1b1ed3c422ed543a2a785fd107278c","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 02:04:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"405b904ab221eb6e456dd888382aaf902c79a38d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-29 02:25:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aa6b74d7ddd6430e828ed37f60e2a41c : SUCCESS in 1h 11m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3cf0e455a4074c27b31c9d8b64c06849 : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9067cd9309627760b5d00d07ec593cf05913ec44","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-29 02:27:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/825/.","accounts_in_message":[],"_revision_number":1},{"id":"ca1f626a62b4db2ec64fb5b484602c15caaf6811","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-29 02:54:09.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/910320/3/check/DellEMC_Unity_FC/788b1a5/DellEMC_Unity_FC/1427 : SUCCESS in 1h 15m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/3/check/DellEMC_Unity_iSCSI/8f45484/DellEMC_Unity_iSCSI/1460 : SUCCESS in 1h 44m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5c8b767f0cba7fccbd5b3fc51b7eda680fa643d2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-29 02:55:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/3/check/DellEMC_PowerStore_iSCSI/6e6fb0a/DellEMC_PowerStore_iSCSI/1130 : SUCCESS in 1h 11m 31s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/910320/3/check/DellEMC_PowerStore_FC/5dae5b6/DellEMC_PowerStore_FC/1292 : ABORTED in 41m 17s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/910320/3/check/DellEMC_PowerStore_NFS/d9b8873/DellEMC_PowerStore_NFS/814 : SUCCESS in 1h 02m 50s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/910320/3/check/DellEMC_PowerStore_NVMeTCP/1746e50/DellEMC_PowerStore_NVMeTCP/682 : SUCCESS in 1h 59m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"3d80721e5793ec71aae73e80137d9b00023ec1c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 03:14:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8be6130fa7d74b95892ed27ae264baa6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/79865fc356e74e86a8f4172612fc4bfa : SUCCESS in 2m 52s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/27c25fa5499b485eab9e21f918eb9ba0 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8b0cd0f80774e2aa6df7e58fb9942e9 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31751018bf2248d4a7e99bdbdf73af8b : SUCCESS in 10m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5c0425525ca40d3a268f10569639a54 : FAILURE in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f05b68b54a4c3d952c5c4783d2020b : SUCCESS in 12m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f2d317a62b144519e035f1d3f0cc4a0 : SUCCESS in 3m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/ffb74af5cc98424496589221234655d6 : SUCCESS in 47m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7040efb2d54c48c9bd37dec18a6f7aae : SUCCESS in 1h 06m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc29cf5c50c4421ea3357bc1320315f8 : SUCCESS in 1h 34m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ca5dea2cbf645f88999b7b5a9564724 : SUCCESS in 47m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af9dcd40733e40feaac34f0a77aba860 : SUCCESS in 11m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b8a594ce0aa342e78d072a0a4ec57ec0 : SUCCESS in 16m 21s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8b0ea640276249e490467de3f2daefd7 : SUCCESS in 12m 00s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9305244d4e224b36ac7c1bcb7cc70554 : SUCCESS in 6m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9a737602fc64209b9f9aacc6613c52a : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a0af8b7e32340cda28ddffccc51aabe : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/715e67ffa1aa4a55855bf0218f9d5b34 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a49a101d453f4709ba45088f98dd1061 : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b55c629c5374a67b1dd3a2b2c7ac9fa : SUCCESS in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/19c4446055444f3eb106f55426b1565d : SUCCESS in 1h 57m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9350a5b7904b4f15b1a2e459d5253043 : FAILURE in 2h 04m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6d5d10b5e0e84cc4aa9341bba5280ec2 : SUCCESS in 1h 35m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/44ec378d03c44dc98ad7c37417f3e4a6 : SUCCESS in 2h 05m 13s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee32493c6f2e4c79b723379349d2d47a : SUCCESS in 47m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f5da824a7875415c95fdc3035e08e5e8 : SUCCESS in 1h 32m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2c45c06d5f824154af0f75113a6aa687 : SUCCESS in 46m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ff95536174bd4f16ae3398ab8e8b7faa : SUCCESS in 1h 24m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4c2d0957ca21481a8011dc00de8817bb : POST_FAILURE in 1h 52m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7493e1d30ea94d47a7954887726342d7 : SUCCESS in 1h 13m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebe15f43bea247daac71ac3ba0c8748f : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":3},{"id":"1a5e6a893fc9d4e5ec86075fe6ae186c68f07e23","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-29 04:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/910320/1/fc : SUCCESS in 1h 12m 33s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/910320/1/iscsi : SUCCESS in 0h 59m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7b0161aeb6086930c7b086b167072da42f329601","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-29 04:00:11.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/910320/2/fc : SUCCESS in 0h 55m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/910320/2/iscsi : SUCCESS in 0h 47m 33s","accounts_in_message":[],"_revision_number":2},{"id":"203b7b27673b9f25cb1d7266225a3c440b7b9905","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-29 04:03:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/20/910320/3/check/fujitsu-eternus-dx-iscsi/4072185 : FAILURE in 19s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/20/910320/3/check/fujitsu-eternus-dx-fc/75992e2 : FAILURE in 3h 05m 59s","accounts_in_message":[],"_revision_number":3},{"id":"fe8e101d54d88ebe71d31fa8968c9cb499365f92","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-29 04:24:14.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/910320/3/check/DellEMC_PowerFlex_3.6/d1495f9/DellEMC_PowerFlex_3.6/1571 : SUCCESS in 1h 35m 55s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/910320/3/check/DellEMC_PowerFlex_4.5/79f14fb/DellEMC_PowerFlex_4.5/525 : SUCCESS in 2h 04m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"6639dd3de8d65b4f6ca2de7b62bd66344f676811","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-29 05:53:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/827/.","accounts_in_message":[],"_revision_number":3},{"id":"d325973198c1ef4d8147788acfb6258d3068a292","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-29 06:19:54.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-1/FC/ : SUCCESS in 2h 36m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-1/iSCSI/ : FAILURE in 5h 33m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c9af02a086d5fcf62b4da0c80bb635e08a553737","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-29 06:29:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-2/FC/ : SUCCESS in 2h 25m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-2/iSCSI/ : SUCCESS in 5h 13m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"8939123ed538e1c869e897fc66b79cf7376e08b5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-29 07:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/910320/3/fc : SUCCESS in 0h 42m 28s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/910320/3/iscsi : SUCCESS in 0h 58m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3605a0d5037b0246b08067617fe51dcb7c989a27","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-29 08:28:56.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-3/FC/ : UNSTABLE in 2m 35s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-3/iSCSI/ : FAILURE in 6h 12m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"7161181aeafc11ec609c0e2567ef8b3dea961e75","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-29 13:42:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e41d27d95e098d51a79fdb0152005ac105bfa14","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-29 13:45:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b2949218e23e7800c75226fc71b562fe5eec4f95","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-02-29 13:49:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"826146f5c6c247ef25a01197a4a67c46a2d05d78","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-29 13:57:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f84b7e093fde32ab7216db447d38787ae0d12989","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 14:07:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10b646c6f10e6112f25b3a66644a198329de4845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 14:33:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"006ad94114e876f00a44d8d0735f128c43d78288","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-29 14:34:03.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16578/","accounts_in_message":[],"_revision_number":4},{"id":"4b4b648dafa870966eb45240be53610d6b2fe97d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-29 14:34:05.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16578/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"20f3281d134d2f822504f7fb9713cccce4a5bbb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 14:55:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fdb73ee95354a10ae7c5581859ed9e4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d5453b8f415430ea6072466de96ce02 : SUCCESS in 19m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9a13ce5ffbc45dd986e266ee85ec6e7 : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/78d10997228349d6873669d00b228be1 : SUCCESS in 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"365adc6bd137cb2c3c598760d70542b4c415c80d","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 15:07:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1beab4bf04fbda34e987c702d14457768640901e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-29 15:33:19.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/20/910320/4/ : FAILURE in 51m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c9d740c364a682efbdf897c37d5bee2dde8bddfd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-29 15:33:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910320 : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":4},{"id":"2dfa54cc2ce4d9ef488b3b675e711a87636e5f76","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 15:47:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a24666c57da8296181c50b657e7ff1c4d9c92e04","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-29 15:59:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8c0e79109a7f4838b14ca8cf639b6ec3 : SUCCESS in 1h 15m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d2eb91a2f82e422aa2a9f37277798587 : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":4},{"id":"4485f16ff32666cb9c291b55234de20b4af79228","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-29 16:12:43.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/910320/4/check/DellEMC_Unity_FC/296b0bb/DellEMC_Unity_FC/1435 : SUCCESS in 1h 21m 46s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/4/check/DellEMC_Unity_iSCSI/a54a9aa/DellEMC_Unity_iSCSI/1467 : SUCCESS in 1h 32m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"49eb48a32f097256fbe07d37b3d9b8438bd91fdc","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 16:20:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04d8e7d31b229f37e0590e402d527277c1fcabc1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-29 16:25:58.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/910320/4/check/DellEMC_PowerFlex_3.6/ed158f5/DellEMC_PowerFlex_3.6/1578 : FAILURE in 1h 44m 54s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/910320/4/check/DellEMC_PowerFlex_4.5/33de592/DellEMC_PowerFlex_4.5/531 : FAILURE in 38m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"20d726a80a517e1ee3695cceca49757b8e0a6048","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-29 16:48:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910320 : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":4},{"id":"0cb592f539dd0d8fb0640705b87a3c0dbc28cd41","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-29 17:18:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910320 : FAILURE in 27m 10s","accounts_in_message":[],"_revision_number":4},{"id":"5cae2dd02be150171daa0e6bd44c9cbc05877833","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 17:36:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9a098b3f9048dfb047d8870ab0f385\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/152ba908b27b456e9ef04ce7f971fd40 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23b357c47dc44c9798439b4d093d286d : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6fb4506adc74570abb81d62bd32ba74 : SUCCESS in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6a961a8562cc5056b8d0da9d5dcadc8f67eb3b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 17:46:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f82ff51ffc5d45fbbe40a093a65a961a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaeed076932e45449e7f04e04b21549f : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/198f3cc772284b0c984f2a1c4383c8f5 : SUCCESS in 11m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ecd9fb8f5afd4423b9c0c62d720f3f19 : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2b0a3eb3eb4837bcbcbaac31131361 : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/4179f2c5ca4142648127311d35e36dcd : SUCCESS in 1h 01m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9788a17f953547ce93bb7589ae2e5e60 : SUCCESS in 1h 05m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b1ac4c5f3e24ac08febcaa6ffaa1569 : SUCCESS in 1h 27m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/186bb4fefc2b4c40ac47040ea6e81a33 : SUCCESS in 59m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4711b21a0f54e3a87e285753987376f : SUCCESS in 10m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8aef523ba5814231be705c85a45df402 : SUCCESS in 14m 35s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d90c5b7910914fff807281059b241496 : SUCCESS in 17m 25s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9566a44e52a84ae495a6c68491125bbe : SUCCESS in 6m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34444569cd774a24af6f399ea15831c8 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d41b0d274a742d0a54b24b94136efac : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13ec4d2c3ba6425c84e09aa718013558 : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ec9ec62ca394d5b833d417b584f957c : SUCCESS in 33m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/698d6134dcdf4994b20d14b2582b1d63 : SUCCESS in 6m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dd3a8a33c8664a70b9e735708b262736 : TIMED_OUT in 3h 03m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eba38602c3714e3bbc927dfad7db2e61 : FAILURE in 2h 21m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de7785136e1e4ad5be1e157c162ce6fa : SUCCESS in 1h 31m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b86dbe1b1afa4e1b9d9ad637acd60480 : SUCCESS in 1h 59m 52s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2e62567ae826409b9fd1bb3ea3ada533 : SUCCESS in 1h 10m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/63d38aa0708b406fb753d61693167551 : FAILURE in 1h 22m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/63c22ac5bf42468883069fe4c8d25280 : SUCCESS in 49m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/444d742f9ae242de97664f458cbc1ea9 : SUCCESS in 1h 24m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/913962ba28574de489d7bfa1861ca699 : SUCCESS in 1h 39m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6de2cdf002d0472d90d9c490dbb5be3a : SUCCESS in 1h 38m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7567bd1ee5d43bbaf078f481bdf1c42 : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":4},{"id":"aae6d48ab30e3e912d8da1a82836293372f105cb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-29 17:48:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fe5cbcf31bd24f4c8f607a9a7a651312 : SUCCESS in 1h 12m 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ec3e711550994f77922951ae3244f4c2 : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":4},{"id":"7863ff83ae1a2070dee730c55f1d7a2d7c3792c4","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 18:03:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9972ba9f7f29ae699ce15cbd0ce768cf3204efa4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-29 18:03:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/833/.","accounts_in_message":[],"_revision_number":4},{"id":"461d2318051f310d9b76f560a671be249bf8452e","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 18:03:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc767a77c6c23c310a9e6ff288674bc8d6574d6a","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-02-29 20:53:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"46b321ab899c42791d43b8dc6f0a60710a84ae44","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-01 09:19:33.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-4/FC/ : FAILURE in 6h 16m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-4/iSCSI/ : FAILURE in 18h 10m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"5d25b776017eac880c6755ebeb5f19b89183b2d7","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-01 15:33:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e310ffede641bf41632649c6581e3c91bf27772d","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-01 16:21:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2d34ef9c9bf2d091cf054afad53b1101ec82832d","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-01 16:26:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d4cd1bb3343cbf5bfb72a9e168ab951e2594f040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-01 20:28:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1e7f7f148eb6afcacf519e056b7f8ae41a5b038f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 20:40:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d60711f1328b42c6b5120a276ccff18b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2d5365d4bc7491e9f8d966e2f2613f9 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/55e80f9d33cf4ca3ace7fc832bbeee01 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/148710aae5bb4083a98162292c45fb5c : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b02e6d0db81067a39b9c8479f591e806429dd762","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-01 20:46:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910320 : FAILURE in 16m 53s","accounts_in_message":[],"_revision_number":5},{"id":"ccd3416a5ad28651583a00b859586670c95976ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-01 21:43:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a3c4ab2711aa4843a7fd27aeb53c6629 : SUCCESS in 1h 11m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fd0d4f9064144a62a9014a3ab8e989e8 : SUCCESS in 1h 13m 40s","accounts_in_message":[],"_revision_number":5},{"id":"a2faf4af278db735eae4858766445f686ca331c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 22:42:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae262f5d9db34581886ec556ce0e24ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5014b5da234efbbe7a981c755ad513 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80d2c068ea5547c19090b76bd906410b : SUCCESS in 8m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/182e15a814724bcf928448498feed6d5 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92d9b8cceba4b2d875f69776e73bb76 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/e4562911d7d44791a3a8c7e4c0fe7d3f : SUCCESS in 1h 07m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3288188545fe4e2484dc0b136b1571cf : SUCCESS in 1h 09m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c301ee45550d44fcb8c636e46890b4da : FAILURE in 1h 22m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5be4ce333a74e14be235af5021d97ee : SUCCESS in 1h 00m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c764416e102f4742bf7d4ee20194746b : SUCCESS in 10m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a8ac5e2eca964508a35eef5da5ad7ade : SUCCESS in 16m 09s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9b76657d62dd43d293579a2b178b5ae2 : SUCCESS in 13m 13s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/42c4be893569423d9adbf4546c7fe4af : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22521efd54f3447ea9b0c54632476679 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27b4bf89a2124a26ab5af355da1f1167 : SUCCESS in 5m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95cc2f9f8f274427b703e3340a25ef69 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c9e56b22b5814329b63971a41b072cb3 : SUCCESS in 51m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d66098e098a844f9aa1a3d55f197d0ac : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d9d2f12d5eb64a94a8897dc633b2772c : SUCCESS in 2h 02m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bcc0998bdf454ea69525120750edba77 : FAILURE in 2h 07m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4204b79ba2864a969a658f65f9ff9b05 : SUCCESS in 1h 40m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dfb98d073a6a4560ba2e97a2db2551b2 : SUCCESS in 1h 55m 05s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d7a11b0e59b84881bbfb1a18e3cfb6a7 : SUCCESS in 1h 18m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09ccce41a44f42159d862497d0f7d611 : SUCCESS in 1h 31m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54a19a9b685b4d12b03bcdd09da2cd2a : SUCCESS in 33m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/176e1f1b685046ceacc2ff82eb4adbf9 : SUCCESS in 1h 04m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6538cb470c9f41ecb65c5002252a5e72 : SUCCESS in 1h 04m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1efa797d034943f8aa5e92b8a8fa0a93 : SUCCESS in 1h 19m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5945e01cc3884d8cb45e35f19ef951c6 : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":5},{"id":"7781d8d6e7b2f8d51f0a64f8dc89a8beb00f3336","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-01 23:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/910320/5/fc : SUCCESS in 0h 31m 23s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/910320/5/iscsi : SUCCESS in 0h 46m 25s","accounts_in_message":[],"_revision_number":5},{"id":"14699da5f797b4b517da15e70d63e2db511374fb","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-01 23:09:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"70e77933f044c559d6054571a519770adafc057c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-01 23:24:21.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_Unity_FC/48ed64e/DellEMC_Unity_FC/1442 : SUCCESS in 1h 11m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_Unity_iSCSI/3e8ff34/DellEMC_Unity_iSCSI/1474 : SUCCESS in 1h 22m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"92da0a2877af30fea6e10baa425c3375956c931d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-01 23:26:24.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/20/910320/5/check/fujitsu-eternus-dx-iscsi/4a408f6 : FAILURE in 2h 55m 35s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/20/910320/5/check/fujitsu-eternus-dx-fc/35f400e : FAILURE in 2h 57m 39s","accounts_in_message":[],"_revision_number":5},{"id":"19509001f382448d3fbc461d74c810812a5a6fd0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-02 00:53:13.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerFlex_3.6/a3173cb/DellEMC_PowerFlex_3.6/1586 : SUCCESS in 1h 36m 40s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerFlex_4.5/e051339/DellEMC_PowerFlex_4.5/539 : SUCCESS in 1h 52m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"8c3a0c0ee5b4088db8a9f2b798749ee430b6676f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-02 13:40:21.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-5/FC/ : FAILURE in 5h 22m 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-910320-5/iSCSI/ : FAILURE in 17h 11m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"9da1b9be7ec9b5770d16fcf383733ca63951fc85","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-02 15:00:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bd6406c36f6f6e84ad894fc488d198bb991460d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 15:34:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c4f420bfc9e45b18aeb2c99fb6415ce\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee23c4ce47a44a38b076a2d7d8d14014 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b9555206da14ec0a0548f9397a23e06 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/23a4eb332014493eae02bc1e94576608 : FAILURE in 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5d5d05f95a9aae73cac730bd484ff431f1bea59a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-02 16:21:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e8c1a5e530ca47ecb0e44af0abbc0d2b : SUCCESS in 1h 18m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f812eef5f2a541bf900a05bd996ff33d : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":5},{"id":"a575ae6c7b91d6cc773a4658a5c4f6c745e4c612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 17:00:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/075b609f5b9a4c029ffb90c714b11800\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ef33e0dc653471fb825ebf888352848 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a03cf6db58e4426b40c5de49d0b08b5 : SUCCESS in 9m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3fdaffa19be40ffb6d87303a30adbb0 : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21df417f346a4cd8b1db99fe9ecb1697 : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/88fd8a1a4af9469bbba06a58839501ff : SUCCESS in 47m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bcfab57d55f44f4f9d821231e33b3e19 : SUCCESS in 50m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05c4e12036bf40f68f2c1c45eae9851f : SUCCESS in 1h 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11145b3f6a514caf95d085010bdca09d : SUCCESS in 51m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba59a4fa72d14228b651a83e44b41e58 : SUCCESS in 8m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1b96acdaaec1426797cea85551fbcfa0 : SUCCESS in 17m 35s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e168c3a61b5642a7b4646277dca725f6 : SUCCESS in 9m 51s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1e42ebf8aa2451185106cb14d8f1377 : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/58b128e0d37e4ab2aa806caa574817f2 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/259101c43c84435ab2c9202510f7db74 : SUCCESS in 5m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1269ac1610574d328e8a50c6639cb465 : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b167452f56274e3ab72be018027f88f8 : SUCCESS in 42m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/108a24edc4634e79a35495afca83c04d : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f20e8e447a0f40868234e455fb6e6369 : SUCCESS in 1h 51m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e192fefbef6345b88109a7a531d12d16 : SUCCESS in 1h 58m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e9c165c41b0647b28a97288d66ba3114 : SUCCESS in 1h 45m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c85eed9030b145d69137932873b88e3b : SUCCESS in 1h 56m 44s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1b98ab94ae07458a823fc9139cb938b5 : SUCCESS in 56m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a389d07e29df40e3b2dc25d6e862034e : SUCCESS in 1h 15m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/30c2b746c6f74cd1941108e84469c086 : SUCCESS in 34m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/edd530d3d01843acbe548abba27f0891 : SUCCESS in 1h 22m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/868fb4d4ba914f2aaada89cf17c0af30 : SUCCESS in 1h 16m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e63fbb8db384405966264255206408e : SUCCESS in 1h 34m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3d32274dad3478796b598c564f09ce9 : SUCCESS in 41m 42s","accounts_in_message":[],"_revision_number":5},{"id":"215acd2f5ba286cabcfec4e5aa91857b2f95ec8c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-02 18:25:40.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_Unity_FC/00590e1/DellEMC_Unity_FC/1443 : SUCCESS in 1h 11m 20s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_Unity_iSCSI/d627e7f/DellEMC_Unity_iSCSI/1475 : SUCCESS in 1h 24m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b5485d590c6dd0a7462b9ed2f082366bb08144b3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-02 18:25:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/910320/5/upstream-check/cinder-tempest-plugin-ontap-fcp/829ad88/ : RETRY_LIMIT in 8m 50s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/910320/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/37e20e3/ : SUCCESS in 44m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/910320/5/upstream-check/cinder-tempest-plugin-ontap-nfs/6ca2e1e/ : SUCCESS in 1h 23m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp cinder-tempest-plugin-ontap-nvme-tcp : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"6a620f7c81677be9377c4b64a486c5184aec5225","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-02 19:19:00.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerFlex_3.6/9399639/DellEMC_PowerFlex_3.6/1587 : SUCCESS in 1h 40m 25s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerFlex_4.5/ce30153/DellEMC_PowerFlex_4.5/540 : SUCCESS in 2h 17m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"8277d991cb069ebe0c1127da77ba22e8915920ef","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-02 19:40:12.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6c39c65611ad415ba55be0ad7b4b4eec\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/dfa8bff31b2e4f018a580a4c9f4c9c9c : RETRY_LIMIT in 1m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8188f4f54bc42cd65560f8f53235ca2759cbd622","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-02 19:52:00.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_iSCSI/87426de/DellEMC_PowerStore_iSCSI/1135 : SUCCESS in 1h 06m 22s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_FC/873b4e2/DellEMC_PowerStore_FC/1302 : FAILURE in 2h 50m 47s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_NFS/50d987e/DellEMC_PowerStore_NFS/819 : SUCCESS in 1h 01m 36s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_NVMeTCP/4e03fc9/DellEMC_PowerStore_NVMeTCP/687 : SUCCESS in 1h 50m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"f1fa0f8889e7267e75ab6a4abb07ab171a3da4c2","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-02 21:35:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aa6315644cec8111b5493eadc51fc43add1b5575","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-03 02:24:25.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d8ae46133fa7cf43b5fb2c5a40ce4c6810b6151a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-03 21:26:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/910320/5/20240304015746/NEC-ISCSI : SUCCESS in 4h 04m 45s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/910320/5/20240304015746/NEC-FC : SUCCESS in 4h 29m 01s","accounts_in_message":[],"_revision_number":5},{"id":"5537054c45d0b3385c2bc9bc03733ed08901dffc","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-04 07:42:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ee7f88af554b17775223576aaa3367d0592278ba","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-04 07:42:56.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"39e4b70d2ca3ca9eabc350a0d48a71e0da1b86f0","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-04 15:25:02.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3754cd38041063f388939144688afaca287d750c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-05 21:40:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/20/910320/5/ : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":5},{"id":"94236d494f50b5e73b2a00fbae9bbd962d6073fc","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-06 10:56:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"536570b818a89837a81264d5156b861993737b33","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-06 14:03:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4144ba3bc539dda4828516d6072f05657f32456e","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-07 00:52:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"edac2ccfd459f090abd9fcb8966c88d45e2c45ba","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-07 02:47:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e31453d94ed44103ddf58a54a4b191fadc9be248","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-07 17:42:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7af6bcf4e87f8dd7fe06e223c6d059bab364faf2","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-07 20:41:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2b11ee0c05a7f0bfacdbf49ab2a1781bb505b0e1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-08 17:34:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e65e02f7f4278401ce71bbedaf5165a19601d4e0","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-08 18:14:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"342343b28518a55fc44112787c7194fef16dbafd","author":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"date":"2024-03-11 09:38:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1732be40a9efaceb0ceefe1e6e3fdae7956137da","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-11 14:42:24.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerMAX_FC/d5fa024/DellEMC_PowerMAX_FC/331 : SUCCESS in 4h 20m 56s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerMAX_iSCSI/37e1dd1/DellEMC_PowerMAX_iSCSI/351 : SUCCESS in 5h 03m 04s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"66ad94ff990ae20750762a108c3613350aaaf93b","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 14:54:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cfb88910af4fa50fecc3b46e0378f844e2be12b6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-11 18:18:16.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"37e9edbdc54af274ca68191ec3ce7d9b535c1f27","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-11 18:20:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5afdd95f16ff162144c7896dcd091b090ed7fdd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 18:21:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"779175f8708cfbdda6d5187feedc2610b90ff325","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 18:28:17.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e14f355a22eb16ea764bdd8184f8b0727b0f7996","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 19:53:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f6685c2b1278627d6f902d3c603f5cee53a6df55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 20:14:36.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eff551ad515c441d995893e71daa8f34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d38ecd90a497494b83d5c3cec38ef3dd : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/255704f5b5d74a5286ec804beacfed5a : SUCCESS in 11m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/847d6f519b5a41ab89afc80bf0e186e4 : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a494721c8a42ed9a0e91a334196ca1 : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/e363448a50e444c3babea9b0058b201d : SUCCESS in 58m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/25e22879af6f4b79b9a35a62ac1bb75d : SUCCESS in 1h 00m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40377708e16f481889bde5fee6b0edcc : SUCCESS in 1h 32m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2241405775894ec080b65c10ca213635 : SUCCESS in 1h 07m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91a816ff619c4979a0d5df4728182cef : FAILURE in 15m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9af693b75b764b1f9c24f9a7773c9c1f : SUCCESS in 1h 12m 36s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91a3eb9ea30549f39af7edabfada1ce2 : SUCCESS in 1h 48m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f677bd156e948c2a82a64a1fd13e83e : FAILURE in 14m 01s","accounts_in_message":[],"_revision_number":5},{"id":"c2810194dc87d05c6e0e1d1d52eabdb4aa785255","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 20:20:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c2136b485008751bdb1724051de425b93c4578da","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 20:21:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"847547678cdeeaabf9c1aa538d68ffd3559f59d2","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 20:44:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"55fd0793571e932a8148ba002240aaf8fda9b880","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 20:45:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d2cda0fd4bbd6e0029edafb76f84f52a50f7bffe","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 21:44:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a047370acc343269278917676a9f1c54a9da0ead","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 22:30:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1d954e7102f7923313ace72cdb3220bbb886dedc","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-11 22:32:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e96dc90221cc8b9b458af6ed9e7052c7732263c5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-11 22:37:33.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/31f4bb3d3ec846939d0cb5282b57decf\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bab83bbaeb174af198b5239fde9494f1 : NODE_FAILURE Node request 200-0000002731 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"37a3d77ac4de3891175b00f3cd544125caffd628","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 22:48:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa7d7143719a4e1ea67e273e8f2a37f9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/704c193c567f4b579e151c968edb148e : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c91829f3c2b6427cbf7785e90fa9f659 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01a8d1786c77493bb6b6e5f8dafdd6df : SUCCESS in 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1a040080dd9770bcf5ec942c1f4a2ae0d2a93661","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-11 23:07:33.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_iSCSI/4bd6f31/DellEMC_PowerStore_iSCSI/1150 : SUCCESS in 1h 13m 24s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_FC/ce1a897/DellEMC_PowerStore_FC/1320 : SUCCESS in 1h 24m 39s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_NFS/bd0740c/DellEMC_PowerStore_NFS/838 : SUCCESS in 1h 01m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_NVMeTCP/c484f7b/DellEMC_PowerStore_NVMeTCP/700 : SUCCESS in 1h 37m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"0a2f985576958f72b9e125a1c5a0da512923bb4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 00:41:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff39337a12734ff1839912bfeed16fa0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c44cb441324a95b64576d0c206b59f : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3148697f9ed64e64af1cfdd62a8008f9 : SUCCESS in 10m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f7f5cd552a14239941d2e24ce5a4b9d : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db2191872af54605ba1e80ee73397164 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/60e6ec5d245748248e72f166485f3689 : SUCCESS in 47m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6cf3a2088f274cac9b261e008a70fa23 : SUCCESS in 47m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a9151a438cee44278df8d7e30dcb7428 : SUCCESS in 1h 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5601d5af7dd483ba4986b522871d41e : SUCCESS in 45m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eca6caa5cee044239a51a0ad228b19a1 : FAILURE in 12m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7797a9216440495b852d1ac7ee7d4066 : SUCCESS in 15m 17s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1690c4d5ada94d0a8816dedaeb59d62d : SUCCESS in 13m 56s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/913d214ac7054f178c0eb15b78dc1283 : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/24d7e477e3f84f8a812bc0b3550e1e06 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8b40a0adc83d4b9c87517437cec3e6aa : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d63dde953d36491db91a6112ff00db89 : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b69583cdc399491f8e461a6febae621e : SUCCESS in 43m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a0fb2a8265ba495d8a88d4cf1996473d : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/88cf17b5337b4069a342b97b05c4a66c : FAILURE in 1h 47m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/19fbc6bd063d474ea8fd25d6fb340c26 : FAILURE in 2h 05m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f1705724cfa4a3d8363ac3f301b781c : SUCCESS in 1h 34m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1e352c26213e4cbba9ff499fe5ee62af : SUCCESS in 1h 25m 15s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/017542d7eba84148b76770456017051a : SUCCESS in 51m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/57627b315e57484b9c9c2a5b450cbd1f : SUCCESS in 1h 16m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/871c108da255416598c4fccec511ee17 : SUCCESS in 46m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4e44ac71cf774ccc835fa9118f587b2b : SUCCESS in 1h 28m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/878e81207e3247d99baa73eacc4f3c00 : SUCCESS in 1h 11m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4e7fe482efc478281b9d325b803a258 : SUCCESS in 1h 16m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a69e170b5b74a9694e49363022e31ae : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":5},{"id":"9c6f2e678c762304290eb526f8ab9b01913c1fc8","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-12 02:35:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e8b07be0e96a0b64d9dc2e9c19e1a86eefcdf469","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 03:01:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36c39ff5877942289e2dff45bebf4d61\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01ae0d252fa048a4a51f74c3f7ef77a3 : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47c9ad2916fd4dae89ed5d7ffe086eda : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5fc829d742e849839053ab579633a4ee : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f693cdc889f5026f00d0310fdf0891dff7aff945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 05:41:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e35c11416bb446e7a934aa2c56b474e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fef5ce972dc4fd2b27c0e1c327009c5 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16ab02606faa4109b281651196897ea4 : SUCCESS in 10m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d795c4788344856a75ad16874a156ae : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca01dcd725424bd19d343936043d792d : SUCCESS in 13m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/746019d887cf481db44af7115baf54ee : SUCCESS in 1h 03m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5e76141cdbd3415f951fa2b06fb2d48d : SUCCESS in 48m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e20d926c529411fb71c9661c1401f54 : SUCCESS in 1h 18m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b66155ac83714e7994e24db3ba8ea4a3 : SUCCESS in 50m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/014a6072f6b544e4829290ba683360b4 : FAILURE in 11m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bc6bad8f20a64ae5b8ff03635d810717 : SUCCESS in 14m 05s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/52fa8a4889db4f88a5506898ec9bc439 : SUCCESS in 13m 03s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0178da931e3c4c5b861e66ab13d2bc3e : SUCCESS in 7m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/37a32248340b475286977dfe625fdc37 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e120f40d94bb434d818fea7bdebede0c : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d9310462baf49b686cd3b8284a2fed2 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1c6377280c564988a4d839d9924af249 : SUCCESS in 42m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1694d196ac9b455ab77d81bab495a69c : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc79b29da25845df81e269eb35d476a0 : SUCCESS in 1h 31m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/792d80eba4994be285d677cd38d38f8a : SUCCESS in 1h 59m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30af233b04b9418a8d35158ffc971f15 : SUCCESS in 2h 28m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/92b7e8982ea34873b7a7bc50858cd1ab : SUCCESS in 1h 58m 53s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2225a5dc6c144f20b4705b2273b2dbe9 : SUCCESS in 1h 05m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ba56e97703b64a70a4fe1c0d07c63f09 : SUCCESS in 1h 29m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a7300788dec14c9982dc7fa507df1f50 : SUCCESS in 44m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d38f63637ee6497fb0440a22a6363b53 : SUCCESS in 1h 12m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b867df0b393a4e37b9417240026784e7 : SUCCESS in 1h 23m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4aaced03d5d247d7919e3d0563447b42 : SUCCESS in 1h 30m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddaac27a64fc4611a438afdd1880fb4f : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":5},{"id":"d8e779a45e992114087c024995f6287c0549e0ad","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-12 06:11:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ea21b56b0ebcfd78ad1b38f8c6cd9de7a6bc6695","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-12 06:58:13.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/cb25e6a27ef942e7971298ebc0d88f2a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/71e518013dfa45858873cd380fef0c98 : NODE_FAILURE Node request 200-0000002735 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"e9ceb59a4928a2bd8e3f35a1a7485af009fc45a6","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-12 15:30:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"080063b49d54654bec2da2b57fff91c646fdf499","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-12 17:43:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b37476826163f703d777c1bfecb99ebccc05ec9f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-12 18:25:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"76f4971b1838f1c9648efb3fac64411305250038","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-12 18:44:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c202f5c13a45414cc8db98e6bf42812f853a03cd","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-12 19:51:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0ac5de036a930e5eb42beec0a46f1805d8b1847d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-14 18:01:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"27aae2a44e1de7569a0600c8684b83c586bd74bf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-14 18:09:23.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8e006c4109d9481eb0b909be1844540e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d76dd248a202426589ea9f528f0dfd18 : NODE_FAILURE Node request 200-0000002775 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"70f97bedd57f6a858e0996a8b56649101a7c4f13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 18:20:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cff8ac0aedd411b997ff86509ccbc2d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35928312c1e3406682df59a3e2335b0b : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d68e659bc90e4d7fb1bba6867ff09209 : SUCCESS in 17m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1edf6f06c8714233a666fe850de92acb : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fb91e3497829e26fd96c8a5afcf5e89a8fac935a","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-14 18:33:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"596a93338c5fe5a3b8bb52e7bf464c70c2b61013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 20:28:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/882dd76f3898470b89acf45733bf311e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0bffd710f4a44c9a2bbbb3f03bbfb31 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e6320fb72df431cbba5bfe88c25460c : SUCCESS in 9m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53d70910843b4991a5800b998c88a45f : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c0db9b730c423da94e760f8954cbe8 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e1d1b4a245044a7580e088083a98e218 : SUCCESS in 59m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3c36442ba286489c9df32cda4991a90f : SUCCESS in 51m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a814b8f8e6740d2acbb858aa4a9d11a : FAILURE in 1h 30m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c22c2d2b9094c8cb9b8261eaf462fdc : SUCCESS in 57m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72044e8b8ef348d68d501348f8ba1527 : SUCCESS in 11m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0fb8b538219d47fc9d0148d78883d663 : SUCCESS in 17m 29s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/deed163dfa7046958e6bf7928b569850 : SUCCESS in 13m 17s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f202788067d40d2b3232aafac178229 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f2b6541d8f94458b6b135f9ea490d42 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1c141ede9a34dc4af326d36cec50503 : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ee565a1eb9974f7fab1e88bfb71a0ffe : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0890e03c339f42c4880612f6d59ff4d7 : SUCCESS in 43m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/24b065dff73946a5b46f76f2e77cf3c5 : SUCCESS in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e74852bb667d4624b16c09959841d936 : SUCCESS in 1h 53m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f67fb26b0c5a46479bac1f4ac1059083 : SUCCESS in 2h 15m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c36134b25f74d59b1fa1965f4180e2f : SUCCESS in 1h 50m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/616dea791d6a4994b24797d0691796f1 : SUCCESS in 2h 04m 51s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b86c8e03e3454b8b869282b3c0bed6dc : SUCCESS in 1h 19m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/78b482d6f1e74e898e142b7696c2043f : SUCCESS in 1h 32m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ccf5cf3a2b0a4241882f9d85fe00e19c : SUCCESS in 40m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b956045a68974da59629ef6dd4039a8d : SUCCESS in 1h 17m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3762da5010194dd9a11c43e594a608fb : SUCCESS in 1h 39m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de64afb434ea4c75a0f566dc11621950 : SUCCESS in 1h 27m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/753108140e8d42528d48a51493e1b227 : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":5},{"id":"c055d92d09d8224c9f2682a7964c6f036a506791","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-15 00:22:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d28ed73b1e31798111c76107e47c072456241b48","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 00:32:28.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/69a299042ec94663b4b053803115fb99\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bf1d27c6c3f949f78fcea00d3282fc75 : NODE_FAILURE Node request 200-0000002776 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"d4e47c87291c52407034764d101426de20de28b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 00:45:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d0e86c00ad847abb69da4f47f45222d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c519b8081a44598a36cdf6719c4b8e5 : FAILURE in 8m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/558d2980d3384a4ba6aa513d303be158 : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52b36820e2504119aa8280951038f339 : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bcef7ab8bd383fd67fa6d766aff9b254645c666a","author":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"date":"2024-03-15 02:22:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"613216e32ffb23e81f5684214986c20620ba8428","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 02:29:34.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5d6d5bbbd740459483112f57572fc736\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6bd58bd4d290465f9375be1d9fdde19b : NODE_FAILURE Node request 200-0000002778 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"4aa7199686548779f1f45fac93171afad835e36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 02:47:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07da6ea4224547578487e3ae034e2cae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1cd3f0beb4f44afae1450e82e5b4430 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33d72965d570458893cb7e574b904275 : SUCCESS in 10m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f759c165931049a3a28671cb439ba836 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761fe6e5e9e342959dd5a0519335d8a4 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/ab2d8efd77ec471a80a018b744aefd34 : SUCCESS in 44m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3126fc9c3ade43f68a85a4ee211ccd46 : SUCCESS in 1h 04m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31ab12f04c964f0f9c153dfb8b3a26c4 : FAILURE in 1h 25m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3635273d03d841b993244c34a246d33c : SUCCESS in 1h 01m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ec9d92cbf734e13b150af3c5114928a : SUCCESS in 12m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/296e53328bcd4eb0b06e547bf13d4e95 : SUCCESS in 16m 33s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5b3b11a41dbf4282b09c5f112abb56db : SUCCESS in 15m 40s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/21db60579fd942b0b17c5b98ea55195f : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/179187166d6a4b519f46fab138e3f2f8 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/150685f34ade427791e09673aba9e484 : SUCCESS in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/417e6385260b477b98ccc16b4c60f885 : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9aef3a8beae64468af274231212e6615 : SUCCESS in 31m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/92b5e30b0eb24a9ebffe8520dccf898f : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16fef7a37c8840c4bdef0401357ef40e : SUCCESS in 1h 52m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2689f88869ff4946abd889aa7563bf04 : FAILURE in 2h 17m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de79fe3e98714ffb8f064376a49e366d : SUCCESS in 1h 40m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9b302d3de0b84a7d8dd411ae62978899 : SUCCESS in 2h 00m 53s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ea6f6a037d0b4954a7d63392fec4954a : SUCCESS in 1h 15m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a5f03c13856b46119b9088e31c7e7f0d : SUCCESS in 1h 33m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7692b7b3e140424ba8bafff04ac8c3dc : SUCCESS in 35m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df2b38cd9abd464ea5d8d12650041bc1 : SUCCESS in 1h 09m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ff147f83f72d43a6ac55f4776bc5dcdd : SUCCESS in 1h 43m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eedd5bfa1e2c469c8deedc9b6af41267 : SUCCESS in 1h 23m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/063e2c9df2fb4274922fe8599055dc14 : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":5},{"id":"c48a21015031ef5b3c8cf5d9db492673b5414fca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 03:01:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa313882676948e39ec3e8c2b1c0b315\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec123d5cab134e9fbad660eb158e0cea : FAILURE in 27m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f59c8d05440d4d3cace1cdb2fad0c12c : FAILURE in 31m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6b9944f4fdfa48cb9124b368a611f5fe : SUCCESS in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dc0d1ec0664f2ee0667fc7892e526e9a9dda5141","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-15 05:13:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf2e129f3037a8cca2338740b4af3ec5c68f7e61","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 05:19:25.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1d6e206c13994b8f9dc8e4d7aa9aaed3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0c1a4ecb7b9242aea581f68ce1bdcd63 : NODE_FAILURE Node request 200-0000002782 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"866724d1608db80ba1dda786320ac345f39d8a51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 05:38:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8d616cc30d400aadc6adbd316dcede\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40449d6c33d947219c48139eabfd2a14 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5ec30a2a6d24365a30f6f578d3ece54 : FAILURE in 22m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ce8c6572a2443bbbaa5ecdf120cfbef : SUCCESS in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6c81c0d038adbcc7e007318c4bb161c017f3d3f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 07:30:57.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c542fd1f6155477c815669a88ebae801\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a54321c27714816830222160c4083e3 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e21c9560c1dc4286b3fad7c6b4919d8b : SUCCESS in 10m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e03b085712f142b3af46786d38f2f6c5 : FAILURE in 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ff4ca331cd43c3bac4a2e76d805d18 : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/1c9b132654f541b79cb6fc9fb9dce3e0 : SUCCESS in 57m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fe318ac93cd249e0a9f34b1f6fe0f01f : SUCCESS in 1h 05m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b46c206b7b4744818a081ad41be8dd83 : FAILURE in 1h 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31d771c770c241c988912ef1b5ce4847 : SUCCESS in 43m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e1779ee9225432381358ff209682a71 : SUCCESS in 13m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d20b620fe4324d2398f4b0bd672056a6 : SUCCESS in 16m 32s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8255a7e8d6694a8c87006bc20d9bce10 : SUCCESS in 10m 40s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f274185c82942df8a3ad6077fe6855a : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9042e71ce6d44441879b0b683298f8a9 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8553a4a76b14413b8e7a86ec27eb181 : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dbcf2d20f53b4d17b3a28fe1aa2c53fa : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/98052893d5d1444a9aee799c5c617279 : SUCCESS in 43m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d5592782b95d488f995d283d08e790a3 : SUCCESS in 5m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/670aa47e1e3f47de83b56016ee0c5ef9 : SUCCESS in 1h 30m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c01081247ba941448127b2e3aaa9b7af : FAILURE in 2h 11m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/244237bce9ad4b80a4aaece1660771ce : SUCCESS in 1h 36m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/84a3e0a3f0c34c7dae7d9f92890f0b7e : SUCCESS in 1h 41m 11s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3bcc21535b2f4b5f87b055acf8414800 : SUCCESS in 1h 09m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/40fdf718d9284ed8a759d047c1394293 : SUCCESS in 1h 35m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a8212d8893b0454ab45a39a2aa3f0dd0 : SUCCESS in 39m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9190bebdb13f47bb84d4278aa0aabd40 : SUCCESS in 1h 15m 25s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3cf8795f7ed14896b4929ad68524e04b : SUCCESS in 1h 12m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9ef2db7d6f14311adf580aa6a103759 : SUCCESS in 1h 30m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4523563a3c34980bf0fa319052d2713 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":5},{"id":"734fe78f82aa33103df98a38fdfb6c61861f9628","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-03-15 07:39:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"efb93bf38a4f6ec8032a5f501afe96d092b8ce4a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 07:45:23.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6fb5065b043640cd904acc44ee1801bf\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1a61cc32dacd43abb2bec2cb67b2455b : NODE_FAILURE Node request 200-0000002786 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"0109c4b75116fd50db7b6b08d97af64f19beb7d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 08:01:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/786ae9313a854151af70d507a0f02cd6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/795954f25353423b88d24769effa8495 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1730397040f46b18558fa77ce2c9ec9 : SUCCESS in 19m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d24aecb184d346cebb3a1e25688c58b4 : SUCCESS in 8m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c1abdd08f45875a505c176fb1acaad784c07602e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:30:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f77660422c240399badd6b98af2c294\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37ad76af8684a379b855caf7d7d0cb9 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/693890ef988f4e3daf1a2804a6b3b86f : SUCCESS in 11m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfdf61fa6510494da7c50b2c4b433f95 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6fb14295b7431ebc0284b02408ef62 : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/228013e4990a4c049fc02a7c2bf55b46 : SUCCESS in 47m 12s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/333b20fe01e149b789060a82838e0c42 : SUCCESS in 41m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18d8326f24194a8e93fcf416b8dd4ebd : SUCCESS in 1h 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6695bdef79dc4d3b80082d1aec9fb474 : SUCCESS in 58m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb4b4d559c05451da254e73eb87d2b91 : SUCCESS in 10m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3836012da7b343d4af38543da845bc59 : SUCCESS in 15m 32s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/017e5ad6d0f94a07b1a577b94734357b : SUCCESS in 12m 04s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b519eb5df7b644959a4e94c58a84aef9 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c1b3f18ca374e17833030494a8cfb87 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/60729496555e4db08d932014f7ae070a : SUCCESS in 5m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/355171f6ce9a4468b22792f8c2af5a42 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8d9095427ea248e18942dcdb2bac0d39 : SUCCESS in 36m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c7dbb73879d413995dca585bb920083 : SUCCESS in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ad27e86ef4064657bfd32883179336c7 : SUCCESS in 2h 08m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9202f30dd8824131b25f754ec9478b62 : FAILURE in 2h 45m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e51fc5023a34fadbb2fe83a9f91fb4f : SUCCESS in 1h 14m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9b78102a643e4115b6cf8a77e2168b46 : FAILURE in 1h 48m 25s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a3dbf0eed7847349c37db26a4918fbd : SUCCESS in 1h 17m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b6ca3490771747f4befca51a035e2649 : SUCCESS in 1h 35m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/339f3d1a2397497788f5a89003a55f61 : SUCCESS in 37m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1d6bc99eb25e422ab70c9495635a5955 : SUCCESS in 1h 23m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2fd51267e78c4985abdfd0b96742c8a9 : SUCCESS in 1h 22m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/30a3cbc8480247448457af56fec8deda : SUCCESS in 1h 19m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10b419503ed24ed99adf9f8d5228b965 : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":5},{"id":"68f91170f646074caa94d9ffe2cb7b8edc20ae67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:31:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cf6ea7c1082d0e18c0d8eb3a993680e75b373f99","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-15 12:04:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/910320/5/upstream-check/cinder-tempest-plugin-ontap-fcp/1718125/ : SUCCESS in 1h 05m 55s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/910320/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/833f058/ : SUCCESS in 47m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/910320/5/upstream-check/cinder-tempest-plugin-ontap-nfs/141ab56/ : SUCCESS in 1h 16m 30s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/910320/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/84af107/ : SUCCESS in 43m 12s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"eb815cda2bd805a0df8a28093d7bf93d9317a8b0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-15 13:37:47.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_iSCSI/db42a57/DellEMC_PowerStore_iSCSI/1164 : SUCCESS in 1h 07m 46s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_FC/be16b24/DellEMC_PowerStore_FC/1334 : SUCCESS in 1h 36m 06s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_NFS/1da4415/DellEMC_PowerStore_NFS/852 : SUCCESS in 57m 53s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerStore_NVMeTCP/3e9d6bc/DellEMC_PowerStore_NVMeTCP/714 : FAILURE in 2h 12m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"7c83cd983ed0bd8eee0e2703852636fc42dd3a66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 15:07:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dda22a957a84a3dbcc9f2c27cfa55f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b80f18e9875e473a8fa0dfdd13e44a35 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/554310b4b0f64bd78ec54edebae9ace5 : SUCCESS in 10m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e14ce2c78c7a47bf90835bc29312702f : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e7f9784f174e938ac520d0f7d6dd15 : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/a7f4213a975c4a649eb413d60670ad5e : SUCCESS in 57m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dd8c31c7c294487fa1f4c6a9006268a6 : SUCCESS in 53m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/838399b3796f456a9bf23e063478129f : SUCCESS in 1h 14m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b120b750b5624209a02afc9d95f5c6f8 : SUCCESS in 1h 00m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa48a31fd8ec458fa4015944e29d499b : SUCCESS in 13m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eeb04d405a1744d49cbb0ef148b389b8 : SUCCESS in 1h 23m 08s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bcc2b064f632446c90b32e139f1fcd26 : SUCCESS in 2h 33m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c6e6e8e7ed848a2b033ed22fb7378ef : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":5},{"id":"9c06403c5ef57aec12e1ec4e2e5bda2a30fbeacb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 15:07:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"1533957d503e1d14ad9f243e9a1f1a62d94c1b36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 15:08:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6afc30f9c144eb7990397af6fb2010e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f07e485fab4ebfa023efae058e1fbf : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/932b57876a3c4bca9c481d6298c580ff : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5},{"id":"5fe3063004c12f08af885a05e7451db160b49e5c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-15 16:46:08.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_Unity_FC/151fbf5/DellEMC_Unity_FC/1513 : SUCCESS in 1h 29m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_Unity_iSCSI/e7e9f6b/DellEMC_Unity_iSCSI/1545 : SUCCESS in 1h 34m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"710bc79d6288792d7a49074a60c29867aaf0a9ec","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-18 03:46:40.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerMAX_FC/0cc4647/DellEMC_PowerMAX_FC/347 : ABORTED in 10h 30m 02s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/20/910320/5/check/DellEMC_PowerMAX_iSCSI/704592f/DellEMC_PowerMAX_iSCSI/367 : ABORTED in 10h 30m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d221db220ae232f5a651fb65d12c05774b92b812","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-28 12:41:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/910320-5/Lenovo-iSCSI/ : SUCCESS in 2:03:15\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/910320-5/Lenovo-FC/ : FAILURE in 3:02:54","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7e3833920130bd0a25f05bef187a1d857a90af5d","revisions":{"d3dd994ebc613c05595f39aed2f6ee0ef3be6496":{"kind":"REWORK","_number":1,"created":"2024-02-29 00:45:28.000000000","uploader":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"ref":"refs/changes/20/910320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/910320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/910320/1"}}},"commit":{"parents":[{"commit":"ee0f0dee90d3c29927854f2852fe4799937eecb3","subject":"Dell EMC: PowerMax - Configurable SRDF snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee0f0dee90d3c29927854f2852fe4799937eecb3"}]}],"author":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 00:45:28.000000000","tz":0},"committer":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 00:45:28.000000000","tz":0},"subject":"Continuation of the original pull request. Original PR reached max # of updates.","message":"Continuation of the original pull request. Original PR reached max # of updates.\n\nChange-Id: Iafa44dcf0e8f46749b5ef37f0b8d341e8253a3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3dd994ebc613c05595f39aed2f6ee0ef3be6496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3dd994ebc613c05595f39aed2f6ee0ef3be6496"}]},"branch":"refs/heads/master"},"685d65915de80c2cef45e122f4ec9e96ce675782":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-29 00:47:06.000000000","uploader":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"ref":"refs/changes/20/910320/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/910320/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/910320/2"}}},"commit":{"parents":[{"commit":"ee0f0dee90d3c29927854f2852fe4799937eecb3","subject":"Dell EMC: PowerMax - Configurable SRDF snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee0f0dee90d3c29927854f2852fe4799937eecb3"}]}],"author":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 00:45:28.000000000","tz":0},"committer":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 00:47:06.000000000","tz":0},"subject":"Dell EMC: PowerMax - Configurable SRDF snapshots","message":"Dell EMC: PowerMax - Configurable SRDF snapshots\n\nAdd volume type extra spec boolean property\n\u0027powermax:disable_protected_snap\u0027. Enabled when set to: `\u0027\u003cis\u003e True\u0027`, `\u0027True\u0027`, `\u0027true\u0027`, or True, otherwise the property is disabled by default.\n\nWhen set to True, snapshot won\u0027t be replicated and won\u0027t\nmatch the source volume type, it will be a regular device\nno matter what.\n\nIf the property is not enabled in the volume type extra_specs, then there is no change from the current behavior.\n\nImplements blueprint powermax-protected-snap-config\n\nChange-Id: Iafa44dcf0e8f46749b5ef37f0b8d341e8253a3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/685d65915de80c2cef45e122f4ec9e96ce675782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/685d65915de80c2cef45e122f4ec9e96ce675782"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f648c2f897d7d57750a52df112c49b7763cf2357":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-29 00:52:55.000000000","uploader":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"ref":"refs/changes/20/910320/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/910320/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/910320/3"}}},"commit":{"parents":[{"commit":"c0382a1e0aa1bd8d5c8861a15b3177f59bd64f24","subject":"Dell EMC: PowerMax - Configurable SRDF snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0382a1e0aa1bd8d5c8861a15b3177f59bd64f24"}]}],"author":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 00:45:28.000000000","tz":0},"committer":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 00:52:55.000000000","tz":0},"subject":"Dell EMC: PowerMax - Configurable SRDF snapshots","message":"Dell EMC: PowerMax - Configurable SRDF snapshots\n\nAdd volume type extra spec boolean property\n\u0027powermax:disable_protected_snap\u0027. Enabled when set to: `\u0027\u003cis\u003e True\u0027`, `\u0027True\u0027`, `\u0027true\u0027`, or True, otherwise the property is disabled by default.\n\nWhen set to True, snapshot won\u0027t be replicated and won\u0027t\nmatch the source volume type, it will be a regular device\nno matter what.\n\nIf the property is not enabled in the volume type extra_specs, then there is no change from the current behavior.\n\nImplements blueprint powermax-protected-snap-config\n\nChange-Id: Iafa44dcf0e8f46749b5ef37f0b8d341e8253a3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f648c2f897d7d57750a52df112c49b7763cf2357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f648c2f897d7d57750a52df112c49b7763cf2357"}]},"branch":"refs/heads/master","description":"Rebase"},"5033b4a76b9e86189220f4d88b2fbc34796e11d4":{"kind":"REWORK","_number":4,"created":"2024-02-29 14:33:51.000000000","uploader":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"ref":"refs/changes/20/910320/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/910320/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/910320/4"}}},"commit":{"parents":[{"commit":"ba424aae9241551433f80c15555f815ba9c9d8ac","subject":"Merge \"Dell PowerFlex: Added timeout into rest API call.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba424aae9241551433f80c15555f815ba9c9d8ac"}]}],"author":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 14:32:58.000000000","tz":-300},"committer":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 14:33:49.000000000","tz":-300},"subject":"Dell EMC: PowerMax - Configurable SRDF snapshots","message":"Dell EMC: PowerMax - Configurable SRDF snapshots\n\nAdd volume type extra spec boolean property\n\u0027powermax:disable_protected_snap\u0027. Enabled when set to: `\u0027\u003cis\u003e True\u0027`,\n`\u0027True\u0027`, `\u0027true\u0027`, or True, otherwise the property is disabled by\ndefault.\n\nWhen set to True, snapshot won\u0027t be replicated and won\u0027t\nmatch the source volume type, it will be a regular device\nno matter what.\n\nIf the property is not enabled in the volume type extra_specs, then\nthere is no change from the current behavior.\n\nImplements blueprint powermax-protected-snap-config\n\nChange-Id: Iafa44dcf0e8f46749b5ef37f0b8d341e8253a3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5033b4a76b9e86189220f4d88b2fbc34796e11d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5033b4a76b9e86189220f4d88b2fbc34796e11d4"}]},"branch":"refs/heads/master"},"7e3833920130bd0a25f05bef187a1d857a90af5d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-03-01 20:28:03.000000000","uploader":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"ref":"refs/changes/20/910320/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/910320/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/910320/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/910320/5"}}},"commit":{"parents":[{"commit":"20fe5b51f6aad2651b460dc43a48d743573ebd02","subject":"Merge \"Remove six from ProphetStor driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20fe5b51f6aad2651b460dc43a48d743573ebd02"}]}],"author":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-02-29 14:32:58.000000000","tz":-300},"committer":{"name":"Bryan Neumann","email":"bryan.neumann@dell.com","date":"2024-03-01 20:28:03.000000000","tz":0},"subject":"Dell EMC: PowerMax - Configurable SRDF snapshots","message":"Dell EMC: PowerMax - Configurable SRDF snapshots\n\nAdd volume type extra spec boolean property\n\u0027powermax:disable_protected_snap\u0027. Enabled when set to: `\u0027\u003cis\u003e True\u0027`,\n`\u0027True\u0027`, `\u0027true\u0027`, or True, otherwise the property is disabled by\ndefault.\n\nWhen set to True, snapshot won\u0027t be replicated and won\u0027t\nmatch the source volume type, it will be a regular device\nno matter what.\n\nIf the property is not enabled in the volume type extra_specs, then\nthere is no change from the current behavior.\n\nImplements blueprint powermax-protected-snap-config\n\nChange-Id: Iafa44dcf0e8f46749b5ef37f0b8d341e8253a3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e3833920130bd0a25f05bef187a1d857a90af5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e3833920130bd0a25f05bef187a1d857a90af5d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
