)]}'
{"id":"openstack%2Fcinder~910483","triplet_id":"openstack%2Fcinder~master~I5380caaac22ffd50a94aebdc56af47bccc5004e3","project":"openstack/cinder","branch":"master","topic":"fujitsu-driver-update","attention_set":{"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2025-03-26 07:01:33.000000000","reason":"Someone else replied on the change"},"37834":{"account":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"last_update":"2025-03-31 06:33:18.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5380caaac22ffd50a94aebdc56af47bccc5004e3","subject":"Fujitsu Driver: Added checking for cascade copy.","status":"NEW","created":"2024-02-28 03:50:42.000000000","updated":"2025-03-31 22:11:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7244ac221b420e062b9cffc6b969dc787a655ce4","_number":910483,"virtual_id_number":910483,"owner":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-31 03:26:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-28 03:50:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-28 05:59:27.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-28 07:48:36.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-28 08:24:45.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-28 08:46:54.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-28 09:36:36.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-28 12:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-28 12:23:12.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-28 13:58:51.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-28 16:04:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-28 21:13:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 11:57:45.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-01 18:38:14.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 06:49:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 09:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-19 09:08:59.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-03-19 11:30:07.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-19 11:44:58.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-03-22 02:57:25.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 07:03:15.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-04-01 00:32:22.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-06-14 12:38:11.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-06-16 00:01:30.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-06-16 06:08:11.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-08-29 06:17:56.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-29 07:44:54.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-03-21 17:53:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-31 01:13:12.000000000","updated_by":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"reviewer":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"state":"CC"},{"updated":"2025-03-31 07:52:15.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"}],"messages":[{"id":"b897d1dcc19270ab9309d1b2c2bd22350b53d6ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-02-28 03:50:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd838313819ab4e9aa3dc34b5cfc2ff3d43d72a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 03:50:51.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16566/","accounts_in_message":[],"_revision_number":1},{"id":"c97dd9420e3afc5969bab6f030c39408edd3efff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 03:50:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16566/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0ddde95a3b35c8b8abd14a73ad44d0c46a7d325d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-28 05:59:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 31m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f00bb724c74868b743af429cd9266110ce1950ac","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-28 07:48:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/83/910483/1/ : FAILURE in 2h 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"35045ce3d1bf4c2c4b322d26b48ca1e211bd88c2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-28 08:24:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/1/check/fujitsu-eternus-dx-iscsi/3d50f38 : SUCCESS in 2h 07m 52s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/1/check/fujitsu-eternus-dx-fc/fcd489c : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"193e881951bdefcc27cab842bd1929fafbaa075e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-28 08:46:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/819/.","accounts_in_message":[],"_revision_number":1},{"id":"0e7000fb71b3417dcbe2d5fd557aa66a028f08fc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-28 09:36:36.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-1/FC/ : UNSTABLE in 2m 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-1/iSCSI/ : FAILURE in 4h 52m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"038a37b4582ffda9d527c5b1430481243d356d72","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-28 12:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/1/fc : SUCCESS in 0h 52m 53s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/1/iscsi : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1d168b1a194d6770f0d6febb12d166f0ec8871ca","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-28 12:23:12.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/83/910483/1/check/DellEMC_Unity_FC/322e1b4/DellEMC_Unity_FC/1416 : SUCCESS in 1h 11m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/1/check/DellEMC_Unity_iSCSI/51e6279/DellEMC_Unity_iSCSI/1449 : SUCCESS in 1h 33m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6a9237208c97e9949b426951ef2b8883d778df30","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 13:58:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 48m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1ae6513cb58cba827d989ca0043690cf47022e54","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-28 14:53:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 49m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0606c81ad8748c71f3a17feb8e5dff074c0cdd3b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-28 16:04:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f5d4357ddce2435bbaf4bce97db4aa62 : SUCCESS in 1h 22m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/244e64e7f85d41b2a0202e9a64acf014 : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e184a41e4f0f64a173faa0d95d5831cd449f6a75","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-28 21:13:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/1/check/DellEMC_PowerFlex_3.6/82f8189/DellEMC_PowerFlex_3.6/1560 : FAILURE in 1h 38m 15s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/83/910483/1/check/DellEMC_PowerFlex_4.5/befd930/DellEMC_PowerFlex_4.5/521 : FAILURE in 2h 00m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"aa414bdf714f82daa64471e333a12da6a5252077","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-29 11:57:45.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7739df96737241fbbd3d724cd7d3c096\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/89be78a751854be695998d052809495c : FAILURE in 1h 25m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ec63a1495a73a952ce00617a74fe0db5b4acf4bb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-01 18:38:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/1/20240301194204/NEC-FC : SUCCESS in 4h 26m 51s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/1/20240301194204/NEC-ISCSI : SUCCESS in 7h 56m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b4a681eef191464220bb4d5ac33cccf9b5b7e0fb","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-18 05:58:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"967f101658b8c9b60a05fa303284acb0c18ab4c7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-18 09:14:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/1/check/fujitsu-eternus-dx-iscsi/1300b0d : SUCCESS in 3h 09m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/1/check/fujitsu-eternus-dx-fc/d30ec71 : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c79a7371599ca06f87a1a9b70f6c89d03467b8cb","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-19 03:53:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e005b4b54c559a761cf79b47be09f697fb5c6e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-19 06:23:24.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"10632c0df236b349e319718a5757efcc52e9d2f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-19 06:23:34.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16669/","accounts_in_message":[],"_revision_number":2},{"id":"54d68e63d8b1f8d5db38b97688841cdeab20d1f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-19 06:23:36.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16669/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"71e93913172480b9aa52293b441067c1b29db28c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-19 06:29:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/1/check/fujitsu-eternus-dx-iscsi/47d0dad : FAILURE in 2h 31m 36s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/1/check/fujitsu-eternus-dx-fc/b45d2bd : SUCCESS in 2h 30m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2c1c1ceb94200bbc3d168121c873b2a6ef007c8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 06:49:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69f29d774b3410a8c2b795e6222b57d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f944f4a9fbcf4cde907f821b890a5126 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f0a5f3e0f734ade939ef4ddce2c378b : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0561ff5a3b91444c81d39ea86729c219 : SUCCESS in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79d8624419ba16c970ed6997de6d9e30b2b233e1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-19 07:11:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d441d92ba3be45efbb886a315384f839 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/21309639f0424af89e715b1d13d4da88 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"9cc40fc28126392b7840d5c4fd36f9f22b40ce9f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-19 07:39:29.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5ebc4aa66ecc432fa1e3c2570fcd6b7f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/be6c30f3a8af48779758fb73f817cb15 : NODE_FAILURE Node request 200-0000002824 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"a33488920b3ce984c9cd4bd97b512bbc90af6eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 09:02:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f57cd8669934a429ed5dadc010d1f4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3d00ed0c88428281c4b510d3738fa8 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/308abc58f4684d29a9e09bd6ef1bc79a : SUCCESS in 11m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41095285926547e2b97a9624cabf32a6 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101df92f28e443f993b236bc83d4f747 : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3bb2894bbd8e485091071057e840decf : SUCCESS in 59m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/62882931ea0e42049bad2bcb303a22e2 : SUCCESS in 1h 03m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80d7ace0e7d34da9b7b2486967dbcec8 : SUCCESS in 1h 16m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f8313d5a50d41b498dbfd48f29521d7 : SUCCESS in 1h 08m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/153105e97fc94d9b86acaec059466752 : SUCCESS in 13m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1ae66de100ab455c8ea8d11d23239518 : SUCCESS in 17m 01s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/466047b150504e5ea68bbbb7e98ec6d1 : SUCCESS in 12m 38s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/67293fc83bef450bb8dde8a9629925f5 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e54bc3c8468341cc828f783ffa3ec59b : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cba7592eade74462b58b8e456a5638dd : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/830a6b0d4d954977a0342b0cbd4c2b21 : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ebf9936f010c412bacc8b25a292c80cd : SUCCESS in 41m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/91b2329151bb4a548d5f7af096004a26 : SUCCESS in 7m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a84cf5672df424b9666937f6cf98b9f : SUCCESS in 2h 24m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/34dd4e79c83a4aaca09b89170e70ba35 : FAILURE in 1h 55m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71f143b048f84203b30c50550e1c4af7 : SUCCESS in 1h 44m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4ce8dbcce79e4c8ea68d1e37f9820979 : FAILURE in 2h 04m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e1d91a96a5034e93a623a156a7915b40 : SUCCESS in 39m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/62f5bb285b704e059d1e4d688012c5ce : SUCCESS in 1h 08m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9b5d4eaf8f7d40778e168e83033c6246 : FAILURE in 1h 43m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/144e925e25404aec83b0b119e3a699f0 : SUCCESS in 43m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/75f0eea3d7d54d828afc74cfc232119b : SUCCESS in 1h 16m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bbe4eba3f02247beb19fc560e7c9c02a : SUCCESS in 1h 26m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d2c7f7d6138741a19aa657be6e829139 : SUCCESS in 1h 30m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa4260cedfcd4ac89fa6333ca736d749 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":2},{"id":"915e6c29326603ff283d092a256f6762a2a002fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-19 09:08:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_03_2024_09_02_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"2dc110a5da10bc5bbbf718c1b654cfe0fc94d567","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-19 11:30:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/2/upstream-check/cinder-tempest-plugin-ontap-fcp/1644c6d/ : FAILURE in 2h 04m 02s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/43cce1c/ : SUCCESS in 44m 33s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/2/upstream-check/cinder-tempest-plugin-ontap-nfs/53e75d3/ : SUCCESS in 1h 48m 29s\n- cinder-tempest-plugin-ontap-nvme-tcp cinder-tempest-plugin-ontap-nvme-tcp : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5a3d07cc13b1d283fab349f0002cd71f7962dedd","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-19 11:44:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/2/check/DellEMC_PowerStore_iSCSI/c8f5247/DellEMC_PowerStore_iSCSI/1175 : SUCCESS in 1h 16m 26s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/83/910483/2/check/DellEMC_PowerStore_FC/01aa2c0/DellEMC_PowerStore_FC/1345 : SUCCESS in 1h 28m 43s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/83/910483/2/check/DellEMC_PowerStore_NFS/642e2b8/DellEMC_PowerStore_NFS/863 : SUCCESS in 1h 00m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/2/check/DellEMC_PowerStore_NVMeTCP/793c571/DellEMC_PowerStore_NVMeTCP/725 : SUCCESS in 1h 36m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"943b523511777e9da17005f215c55fa93b10ae23","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-19 13:49:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 1h 15m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0ed79de40ec0799628125bc370939f27320d2606","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-19 15:43:03.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/89/.","accounts_in_message":[],"_revision_number":2},{"id":"778a02a658aae495e1b8aa4e82a18b429cac6fd1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-19 20:48:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/83/910483/2/check/DellEMC_Unity_FC/407736d/DellEMC_Unity_FC/1541 : SUCCESS in 1h 24m 07s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/2/check/DellEMC_Unity_iSCSI/55132c0/DellEMC_Unity_iSCSI/1573 : SUCCESS in 1h 39m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"6dcb5c4c5895c0b4e24ca8aedbf56076815e9c19","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-19 20:57:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 50m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f090881dac843ba539f3ac1d19e84cccf9e1b58b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-19 21:54:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 52m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d702dfeee4061930eb5b9d59783749b03f85dbd1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-19 22:00:06.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/2/fc : SUCCESS in 0h 41m 26s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/2/iscsi : SUCCESS in 0h 49m 42s","accounts_in_message":[],"_revision_number":2},{"id":"40f74d8aea63ddd1ec2ba80a886e7a5c562d6f60","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 02:57:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/83/910483/2/check/DellEMC_PowerMAX_FC/07496bc/DellEMC_PowerMAX_FC/357 : ABORTED in 10h 30m 01s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/2/check/DellEMC_PowerMAX_iSCSI/bed9d48/DellEMC_PowerMAX_iSCSI/377 : ABORTED in 10h 30m 02s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"f894602d4f0f9c66da196c47ae075262be0dfea2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-22 07:03:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/2/20240322123758/NEC-V-ISCSI : SUCCESS in 3h 00m 28s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/2/20240322123758/NEC-V-FC : SUCCESS in 3h 25m 14s","accounts_in_message":[],"_revision_number":2},{"id":"774f64eb367b07f29d30adaf78832fa8456ca600","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-24 06:49:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/2/20240324110744/NEC-ISCSI : SUCCESS in 4h 13m 05s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/2/20240324110744/NEC-FC : SUCCESS in 4h 41m 56s","accounts_in_message":[],"_revision_number":2},{"id":"759bcbb1cfe28391f84b7445b90006d39c82709e","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-25 05:46:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14afb3519223ab7f0a3808da3fc434209e6d66b2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-25 08:20:15.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/2/check/fujitsu-eternus-dx-iscsi/d1e4750 : SUCCESS in 2h 28m 16s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/2/check/fujitsu-eternus-dx-fc/b9a6fdc : SUCCESS in 2h 26m 06s","accounts_in_message":[],"_revision_number":2},{"id":"82bf5b2a6dffdca18a761dda8f031a5584d27722","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-04-01 00:32:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/910483-2/Lenovo-iSCSI/ : SUCCESS in 2:06:03\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/910483-2/Lenovo-FC/ : FAILURE in 3:03:04","accounts_in_message":[],"_revision_number":2},{"id":"2137cbc1893fdf080346e4811e12fbbf02251d5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-16 02:49:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0f81bcd6f7a1892765f4560c1abeae76a454712c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-05-16 02:56:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/3/check/fujitsu-eternus-dx-iscsi/cc1822e : FAILURE in 2m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/3/check/fujitsu-eternus-dx-fc/7e72d8f : FAILURE in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"8ba7a6860e689f631308292dba192c1eb414ddf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 03:12:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d737f8a3554cfaa1aa21b228a188ba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc2f0daf1ecc41aea3ddacbc3925b535 : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7ac85e249214069a40b13c4b8780309 : SUCCESS in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fe734ca19719c5e8c8ce6ca2f1a58af2e46c75c3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-16 03:51:44.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 23m 15s","accounts_in_message":[],"_revision_number":3},{"id":"1924929e165b9f5fc1e0698a336d3a0c038efbe4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-16 04:24:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 35m 13s","accounts_in_message":[],"_revision_number":3},{"id":"35fa2d863d0f078b79890c97f59ca325bbdc0c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 04:47:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a979a71eb9dc47f9af2f12ef053cd149\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd5cef6fec240c4b28937ebada0dd91 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f508478fb2304e4583fd2cdcf64b630b : SUCCESS in 11m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07361d7e8b684858a61f4a3d781b1587 : SUCCESS in 12m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b8098a524384f638c90115cf4a85d28 : FAILURE in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d869b17318d43128ce5e65ca773c823 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/5812fd2494244e0aad12546643d082d9 : SUCCESS in 1h 03m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5183983bd0f84f659415699086876147 : SUCCESS in 1h 33m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbfbe63d4cd948caa207c080db94b198 : SUCCESS in 56m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/376509756f4d45cba9059a774154ef53 : SUCCESS in 12m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d445994df68f4a3f866ca33e6a0496ea : SUCCESS in 16m 01s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0e991a832db9497da7e5ac68829675a4 : SUCCESS in 13m 13s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0e1d21c2e2f24d6b8108121f0912dccc : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/330251094f7140f2a8e299ef6061d482 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13b8c0d40ef049b297821f44b8232c86 : SUCCESS in 6m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e05b6aa9c84848d19a0523d32eebcfe7 : SUCCESS in 8m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1adbdbb0df9644a987fcc9769467d3ed : FAILURE in 43m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1c4ef7e78d004f3580d828d07c45ffc8 : FAILURE in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7198c5e43bd9436c8e7ebbb1aae3ad0c : FAILURE in 1h 26m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d52588ba770543b2999dd50a9caeadb4 : SUCCESS in 1h 51m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5b837f6adb8c4f14811675d14b1861b5 : SUCCESS in 1h 27m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/06ea72eb656b4dde93c3edb1e5f9131a : SUCCESS in 1h 33m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a7ef1ba5796f4b0d9e9128c922c6c827 : SUCCESS in 28m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91a57660f1c646e991797ffa4da69784 : SUCCESS in 1h 00m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8ed08dd053e64f01a4a43924d7bb176c : SUCCESS in 1h 09m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0b11e23eb4ec4da2a8b111351640c3d9 : SUCCESS in 33m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a2c18d5540e84b7fa9211a8f9b1a9c90 : SUCCESS in 1h 03m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d7a73a4952794dc08fe6d59e489b5003 : SUCCESS in 1h 32m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a53581b80344e56975a17ece1b9c1a0 : FAILURE in 1h 33m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aaceac781742402e8b465293194a0c55 : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":3},{"id":"4da8b7639859bf6ffd1785aa3c520c3c90b37912","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-16 04:58:05.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/043802f31c78431fbe99a65cea0e1dd1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b1c0e3127e7c444896ea6a007c726fd7 : POST_FAILURE in 29m 09s","accounts_in_message":[],"_revision_number":3},{"id":"3016dc394b9f9f47344fc6f74220ed9ecb4971df","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-16 05:33:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2bdb3038d2d3c3b1d855ba32014586f8e53cba88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 05:45:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e537d8869cea42f2baeeb36d2062419f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0dd765b6cc674c85a09ef37f0b3e7a5e : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a20a1d89ed94f238e71a1f113f05cac : SUCCESS in 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63b5ea99c1d33815fde730de339f05a17dee8e92","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-16 06:07:23.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d699943448fa4c55a46e54e4abfa878d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3fd468b9f16c453b9aa6ad1f4391bc60 : POST_FAILURE in 22m 56s","accounts_in_message":[],"_revision_number":3},{"id":"7e78cd9679b512c3aa79079e74a26ae2dd05531c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-16 06:30:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/396/.","accounts_in_message":[],"_revision_number":3},{"id":"e3c197fee0ae193819e1f3e46f68368578849c48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 07:45:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d47ef8c085742079cd90efb5c10270a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06cd49632ec6418085aa735bb9282cc4 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d76011c5c8004445a00741f3bed01ffa : SUCCESS in 12m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbd276cdc11446308806bbef6cbb5bd6 : SUCCESS in 8m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17d24409dc3b4cb793e828e9f83bc6d3 : FAILURE in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eeb70daaad044f6893fc92f99acda43 : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/796bc504f5d2485bb8d139c90ab80583 : SUCCESS in 43m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/babf3653bd504f54a8e568e1f73bf289 : SUCCESS in 1h 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9782d955b38e4c4dad7a7dedaf11af82 : SUCCESS in 1h 04m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18b95834a7314a108f61712f70949d9a : SUCCESS in 13m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/53048d23970c4b128073b2c65c7eec19 : SUCCESS in 16m 10s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b8bacd0b76cc4cd6b69072fac13771a4 : SUCCESS in 12m 25s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a10dcb252b994557b32a4fdc5da6d69e : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba163dcf0fd44a2487b3455f140f7be2 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c7003c91c82f4302b8d15ecba2a4ee8d : SUCCESS in 6m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7ebc49a534d740c9a6deebf2742a1cc6 : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f0440e81fbb34bc5b0ad849c2538dcf4 : FAILURE in 46m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fdb38b305e8e4c0e8a0069091536497a : FAILURE in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5453101e6e664c3594cd4d7036278581 : SUCCESS in 2h 05m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/471ea5f0ff8b437bbf6bca3429acf4b0 : FAILURE in 1h 43m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7ed4f8a4dfdf49e9ade452bd2245e198 : SUCCESS in 1h 19m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e51dc1aaf40a4a7ebde844bd41362c60 : SUCCESS in 1h 31m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/379d221e38a34b13ad081de3cf8463ab : SUCCESS in 25m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fb5f2d04ebfb4de4bb8f00376bd698e1 : SUCCESS in 58m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c035bc9f28c34b6e9fd0eaadc42451f7 : SUCCESS in 1h 24m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e720bea137c340c490ca8c4d3916542e : SUCCESS in 44m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1afd6636e9884fec8f8a0e8916058e91 : SUCCESS in 1h 11m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9bf5f7d7241e423eb5a86e790f6c908b : SUCCESS in 1h 15m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4ec9c008354f48e79809a194cc131743 : SUCCESS in 1h 07m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1a9cabcf81c48548e1a8c9f198a9628 : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":3},{"id":"accc097cea83db371b76b02c18421042bc8cb0f9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-16 08:07:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/3/upstream-check/cinder-tempest-plugin-ontap-fcp/8265b7e/ : RETRY_LIMIT in 4m 11s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/39e22b8/ : RETRY_LIMIT in 4m 10s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/3/upstream-check/cinder-tempest-plugin-ontap-nfs/8906bee/ : RETRY_LIMIT in 4m 06s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/27ec971/ : RETRY_LIMIT in 3m 56s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"123e7709b752cfa0d9e4bd25b00c5452c0497c2b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-16 11:48:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/83/910483/3/check/DellEMC_Unity_FC/fcb2533/DellEMC_Unity_FC/1870 : FAILURE in 2h 04m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/3/check/DellEMC_Unity_iSCSI/e90caf5/DellEMC_Unity_iSCSI/1898 : SUCCESS in 1h 49m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2cf21426890b4c8ac5d7b280a4feb88dd77553d0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-16 12:05:48.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-3/FC/ : FAILURE in 8h 40m 20s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-3/iSCSI/ : SUCCESS in 3h 1m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"c5a26aa62f3c4b7bab5442db07a9fb5dbb0f98e8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-16 13:58:06.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/3/check/DellEMC_PowerFlex_3.6/3b6f409/DellEMC_PowerFlex_3.6/1369 : SUCCESS in 1h 43m 18s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/83/910483/3/check/DellEMC_PowerFlex_4.5/95d9943/DellEMC_PowerFlex_4.5/289 : SUCCESS in 2h 31m 50s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/6de6f84/DellEMC_PowerFlex_4.x_NVMeTCP/234 : FAILURE in 2h 54m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"2526e5f3d9c4bd140f7f774de5be89c4188bc549","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-16 20:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/3/fc : SUCCESS in 0h 36m 54s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/3/iscsi : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":3},{"id":"59c4f94d933deddb70dd485d63d25ebcb05715a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-17 07:06:22.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c6fda0a6d62f3933aec4a1cc7f26077d16d3feb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 07:22:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/424385c117d74dbaa4dcb3dd368c5279\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5639bc471b6e4039af91e6bba27c92ee : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/87b4f48bf4724b9f998139847405a716 : SUCCESS in 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e9b2cf798f66f10b380b7d2de9a8c1512d3d3f63","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-17 07:58:18.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 23m 00s","accounts_in_message":[],"_revision_number":4},{"id":"08db831141329ae1a83995e5445b8289a6e2b941","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-17 08:01:57.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/941636c4fe664a23a3413101b73397a5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ffc713f305874e9da2b1d2fa68ce4e4a : POST_FAILURE in 23m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3f5a3e3505e1f788e9622ba5573ce5f0dfb2ed29","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-17 08:23:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 34m 55s","accounts_in_message":[],"_revision_number":4},{"id":"9807099a5786b7eb0bfe17871bbe5dda97da824d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 09:16:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f86868a7e8c4f23801ff254f98b2bcb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad96544e40549bcbb3187cb50373c6a : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99f6dfd7649d4c0088333723da382301 : SUCCESS in 11m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10a7013071494b5d9ca8fc3d17ee5f3a : SUCCESS in 11m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e23d4f758254d79b5901818b86bea3d : FAILURE in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510a26a85a314072b27e3201a1eaac47 : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/5445683bd1fc4165b6392fbeafb9ad94 : SUCCESS in 1h 07m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49cf49ce638845d1945a4328f3f68d2a : SUCCESS in 1h 19m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17148bb522b643989cc2e614d180f9db : SUCCESS in 48m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15c10f71607a42eea917842e452115df : SUCCESS in 11m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f2760762dfc49c9b703c6a7ac552dd4 : SUCCESS in 15m 37s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/20155b3c96844720ad97bc5f17f93734 : SUCCESS in 11m 00s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0badc44e87614a0c92cfc95b77f68655 : SUCCESS in 6m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d66f790d04642cb82a331f5354026ca : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/916fc9d77ed84e7da39afc28e632af21 : SUCCESS in 6m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6ae59bafc3f940ffaaecddc95680ea1f : SUCCESS in 7m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce70c70c15b04d528b8c625d20f19d77 : FAILURE in 45m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d4bc984fb9cf4ae5b103d57832a8c0ac : FAILURE in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3b94b7389fb044aabc712218b887ceff : SUCCESS in 1h 56m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7253fd12077741a38fc70dc36d6a472f : FAILURE in 1h 57m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cc06e62c244e46678b82d1a512c325b6 : SUCCESS in 1h 29m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7242b39723eb4768897e0de1f8e90558 : SUCCESS in 2h 02m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a87d183616d64091ac4a15af4c5f7bb3 : SUCCESS in 34m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/377ceafeae164dd5be5459af9e0384ad : SUCCESS in 1h 12m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/457be826c28d4db0ba132d8c5e403943 : SUCCESS in 1h 24m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bd3d9a487ee84e8a878deeb0c58d2593 : SUCCESS in 33m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c9a9198ba8d4a4b8a85509e8db9a9c9 : SUCCESS in 1h 24m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/04e2e5cb6d8b4a52ba5d1119f3d30a2f : SUCCESS in 1h 15m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1732dd128c26438d96990b7b597abcfb : SUCCESS in 1h 26m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c86595d4e3ae4fbc88adee86d0f2c539 : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":4},{"id":"79df6d5baadbf6164603a9f2ef535d0a0fd247cc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-17 09:36:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/4/upstream-check/cinder-tempest-plugin-ontap-fcp/e9641f3/ : RETRY_LIMIT in 4m 25s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/9be7412/ : RETRY_LIMIT in 3m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/4/upstream-check/cinder-tempest-plugin-ontap-nfs/ff0c6b8/ : RETRY_LIMIT in 3m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3db2732/ : RETRY_LIMIT in 4m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"5de8918926d2693f6fdd0d10480fea832ad9d2f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-17 10:11:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/406/.","accounts_in_message":[],"_revision_number":4},{"id":"7e5a8461be3b7a001773f6e305b91fd5f1cfd32b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-17 13:23:33.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/83/910483/4/check/DellEMC_Unity_FC/1ac4cc6/DellEMC_Unity_FC/1878 : FAILURE in 1h 38m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/4/check/DellEMC_Unity_iSCSI/d0956d8/DellEMC_Unity_iSCSI/1906 : FAILURE in 1h 55m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"69e1c1890e4c19c87c73f8a0d0874716cffbc796","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-17 15:09:02.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/4/check/DellEMC_PowerFlex_3.6/e939870/DellEMC_PowerFlex_3.6/1377 : SUCCESS in 1h 45m 36s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/83/910483/4/check/DellEMC_PowerFlex_4.5/4016dd9/DellEMC_PowerFlex_4.5/297 : FAILURE in 2h 11m 02s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/b9fb9b9/DellEMC_PowerFlex_4.x_NVMeTCP/242 : FAILURE in 2h 56m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"4ea1576d4d9f1075c6814fe0c7537a3efbaffbbf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-17 15:28:00.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-4/FC/ : FAILURE in 7h 52m 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-4/iSCSI/ : FAILURE in 5h 6m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"7bd74e9ebf16f4ab281f37673749d4d0f2449afc","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-19 21:10:47.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/4/20240519221139/NEC-V-ISCSI : FAILURE in 7h 25m 02s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/4/20240519221139/NEC-V-FC : FAILURE in 7h 59m 04s","accounts_in_message":[],"_revision_number":4},{"id":"2b04dcb69c459f452782e9405bfeb3a7cca54613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-24 09:55:46.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"76b0d39c7bae004f2ebbd4272df36ff11f476636","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 10:08:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9cf51cbc4a6405694e53d14e386f9f5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/331e3c05d438435eb88f8d601832451b : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6279cbdf0d3e495ebc0c28d9041384a8 : FAILURE in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"507353ffa6d66280fb108fb1d08d6c7fc08ed72d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-24 11:40:25.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3e1551bc57a2468989d85fe8efb227c0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a90071964a8646fc94458fe302ab2eeb : POST_FAILURE in 24m 18s","accounts_in_message":[],"_revision_number":5},{"id":"b5a77f6f2460b416cd363bc08555cea964438369","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-24 12:20:40.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 22m 40s","accounts_in_message":[],"_revision_number":5},{"id":"b0d7dd697ce6ceb946e27d9bbe388919b1586c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:32:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d4d9a60ca52467999fab20cd6f20b33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5d658b793c4fe7a8c4b20c1d963914 : FAILURE in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a4008a9fc684cabb5109d3bafd6a9ad : SUCCESS in 9m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21ae1b1962674a19a5676a4a8473108f : SUCCESS in 12m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdeb9687a49746f6b439daa3d0bb4614 : SUCCESS in 15m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f9c9d4f9e84a02b5c5f01b9a881c83 : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/2d11a7caba04471bacf8744b2b9a9a8c : SUCCESS in 57m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/200d2aece77a4b05a31db1044aee0daf : SUCCESS in 1h 24m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9af111c093f4342ab3581f8090cd95e : SUCCESS in 57m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be5eb80fae4742bfa9c4372fded1dbf7 : SUCCESS in 12m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d4f43508b18e46ada5ffd27b63eb04e6 : SUCCESS in 17m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aeaf841c56f14e03b98af642d3c236fe : SUCCESS in 6m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/458bac313b3843759ead6046fd2fbbbe : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/55d357a34b45410ca8c1f54afabf054a : SUCCESS in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0b52de84bd444731a1cb67bd3f1590db : SUCCESS in 8m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/642ea5a3170a4290976f0327bc898a63 : FAILURE in 35m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ce2e5dde863c45dcb5495029482e68ab : FAILURE in 7m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a714e4f4248a4a4091f243ed847bda0b : SUCCESS in 2h 10m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ae0f3c405701474ca0cccc233f144bbc : SUCCESS in 2h 30m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cc1a904ddb704c9b8735ad5edefced00 : SUCCESS in 1h 43m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a4d45e4da7ff40309df41cc24bcb6e6b : SUCCESS in 1h 52m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0f40829b32b94eb1bc72782d775613ec : SUCCESS in 36m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f63a9a2e72f842188da751f9a384e4b3 : SUCCESS in 1h 13m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8f5e9ae0dde9412b8cecbcd21a8fc713 : SUCCESS in 1h 31m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3c371f2bfa7b4a82b1c98f9d3e3c9d2d : SUCCESS in 35m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cf79614433494be4ab421904427502ad : SUCCESS in 1h 30m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/71cb1a47fb524ae68500215284afcc58 : SUCCESS in 1h 30m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3adaa8ae5bf40119bb990273a54efb9 : SUCCESS in 1h 37m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bab05d130aa43308967498956aff726 : SUCCESS in 58m 10s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/fujitsu/eternus_dx/eternus_dx_cli.py","accounts_in_message":[],"_revision_number":5},{"id":"daa6cac9b8c6af49e681126196327442ee3a7129","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-24 13:29:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 34m 37s","accounts_in_message":[],"_revision_number":5},{"id":"6217dfcac048c7e63e9b204abbcacbc74dbb45d9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-24 19:12:38.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/490/.","accounts_in_message":[],"_revision_number":5},{"id":"1f0aeec9dc0dd33b983f11b5882c891dfc49d0c5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-25 08:01:18.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/5/check/DellEMC_PowerFlex_3.6/73d5edf/DellEMC_PowerFlex_3.6/1444 : SUCCESS in 2h 04m 08s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/83/910483/5/check/DellEMC_PowerFlex_4.5/30f5190/DellEMC_PowerFlex_4.5/358 : SUCCESS in 2h 00m 34s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/faed1a7/DellEMC_PowerFlex_4.x_NVMeTCP/298 : FAILURE in 2h 55m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"52268e3431afd0d2e1cfe7a44c5f07368736c838","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-25 17:28:18.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/83/910483/5/check/DellEMC_Unity_FC/fb4de66/DellEMC_Unity_FC/1963 : FAILURE in 1h 22m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/5/check/DellEMC_Unity_iSCSI/26fbd71/DellEMC_Unity_iSCSI/1988 : FAILURE in 1h 45m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"f3220f1461cbdf034c33523ce2a19ab47bc3111a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-27 08:52:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5e94e43b90f7390903cdb2a3ba7e76306bbd027a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 09:14:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d590ba89c0425485e64846b28b1525\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22f962fa710f4fdfa388c1fd7f892030 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb638682cebd4e9996a0989d35cf24e1 : SUCCESS in 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"05ead570d70e70cd8c58b9d9f24b64fdb157a962","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-27 09:56:44.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/307e61597ef94783b189f8f0f5ec85f5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f39ed450c9fb45d2bbc78879f2343b3d : POST_FAILURE in 28m 14s","accounts_in_message":[],"_revision_number":6},{"id":"ac9971b5c6d672f91812d3342efffcca2669704e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 11:32:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6426f51c75874a16a6bbd3977c7a639d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca19644ec4bb490e9205ebb66a0a814d : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c75ce20778e04e5aafb18ebaf1295950 : SUCCESS in 14m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df0c1150848a45919da05430421c0f3f : SUCCESS in 10m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/998e7f973e824a4089ba67866dc30e6e : SUCCESS in 17m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae46a0321814d999f760a5b8383926e : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/11d72e56e97a4cf8ae808e7d06bd2d73 : SUCCESS in 1h 02m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e08712850ed44dbba6af1de23de9830 : FAILURE in 1h 15m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac4d3397ce8c462e80980e6acee83f82 : SUCCESS in 54m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d20ab3bdd0f744b4a3b3737cef592f22 : SUCCESS in 19m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ac2c2ae55fa45ad9596e2c5bc343caf : SUCCESS in 15m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/08ab57d3de8e4e30a73036c186d4f295 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2320cb1a2714eeaa1841177d95b3071 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ea212e9f64643d4b7c0db8ed8f6ab96 : SUCCESS in 6m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3844d5d0f9cb4b9bb594b1630e69edfd : SUCCESS in 8m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/97a7a2e0fbd84436839ca3ba5f0ff852 : FAILURE in 36m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8b4753101d94a209151019632670876 : FAILURE in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dac1faa774e349999069904539c79ff8 : SUCCESS in 2h 03m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5b96862b88cc4f43b9fdec7a33294a04 : FAILURE in 2h 33m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aba5b2befc244054923afafbd5e330ac : SUCCESS in 2h 00m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a9eecbcf43084c2b82bd80bc7b4c41b4 : SUCCESS in 1h 52m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c92e1309eaa547c2b566cb169444d7fb : SUCCESS in 33m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e7fd3e5dbc5f4758967a17713d219e30 : SUCCESS in 1h 08m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/92e65f2b657c4f17b6c9032be067c5c2 : SUCCESS in 1h 36m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a6a24c7928dd4186a07a08188caf2379 : SUCCESS in 46m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/38c742d0d9164ff4912456bb0b914d3a : SUCCESS in 1h 28m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ef2367c4d77a41678445d03b94555ea6 : SUCCESS in 1h 32m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a7246bf10944ee4a604a6be0273007d : SUCCESS in 1h 41m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54c2b32cf60d477dafbebb7b6e1caa1b : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":6},{"id":"9d50eab214bff4661d9db70475b109d16729aff9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-27 13:30:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 34m 29s","accounts_in_message":[],"_revision_number":6},{"id":"67d2c9f663d5c8bbead3d2942a50f3e700e59041","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-27 16:28:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/511/.","accounts_in_message":[],"_revision_number":6},{"id":"f0298f1123f649095d8aebf7a95986d4026937a4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-27 21:58:59.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_Unity_FC/4b9ed79/DellEMC_Unity_FC/1984 : FAILURE in 2h 34m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_Unity_iSCSI/b2494ee/DellEMC_Unity_iSCSI/2009 : FAILURE in 2h 18m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"f86031ec5debc224a17e57eb963812c3f633fe62","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-28 02:39:40.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerFlex_3.6/1d7bb3c/DellEMC_PowerFlex_3.6/1465 : SUCCESS in 1h 58m 08s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerFlex_4.5/015e1b1/DellEMC_PowerFlex_4.5/379 : SUCCESS in 2h 20m 31s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/e62e731/DellEMC_PowerFlex_4.x_NVMeTCP/319 : FAILURE in 2h 55m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"5ce0ba69c912dd7b5a95c6bb7bb2a01aef057a7d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-06-05 20:50:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/6/20240605215038/NEC-ISCSI : SUCCESS in 5h 04m 59s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/6/20240605215038/NEC-FC : SUCCESS in 7h 59m 44s","accounts_in_message":[],"_revision_number":6},{"id":"e9c3e08ad8c08358dc0fd5d1824399630dcee42b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-06-06 00:00:08.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/6/fc : SUCCESS in 0h 42m 03s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/6/iscsi : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":6},{"id":"37d4796d8a339a4227ce7acfd934e93ecf9425f1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-06-06 21:12:02.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/6/20240606221632/NEC-V-FC : SUCCESS in 3h 41m 32s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/6/20240606221632/NEC-V-ISCSI : FAILURE in 7h 55m 27s","accounts_in_message":[],"_revision_number":6},{"id":"4a1efa3a19e7a3db04646b8d7b5675f80f12736d","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-13 06:08:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a6ed5b550b3c4c7f00d88cb44314f3dc1272c4f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 06:32:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6904d39904f438397501736c598a8c7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecd4fe4beb424229af1672a21f600469 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f1d42066818542839168da1e9a7435d6 : SUCCESS in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"61f19a6c2ba8328dda48a9d42aed75a8362480ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 09:13:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83e492a78bac4754b4d84c21dbc1326a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63c6e62bfef842d7ade3ebe36d45ec85 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf01358f2e6b4a7089fa35284771915e : SUCCESS in 10m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93fba4e0163c4fb0a37ba6cf4c2a5976 : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b6887c432454dae9ec76e4022dd9498 : SUCCESS in 16m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1390828cbd44d88b98ea68468417c9a : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a274750354f34860b791d737ef4846a2 : SUCCESS in 1h 08m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aa787b54efc04f41bb29bb54c0ca0051 : SUCCESS in 1h 30m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46fc275afffc4cb79af0ca700fec062a : SUCCESS in 1h 07m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e4cb0e49b17497dbd34704e43175e96 : SUCCESS in 17m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba94a211233b49c89cea13f65ace923c : SUCCESS in 16m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d9dfea128e4345f9929cfacd74bced4a : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/46eff6d8849a463698600472d0f7f353 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/835449d861504cd6ab272e5e1d8915ac : SUCCESS in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/67b7617ee4974b9aa6a226c0958fca99 : SUCCESS in 8m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/312ca08196244ecca40d45ad1c4c166b : FAILURE in 39m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/25f9502c719a441b9932b5467ecee3d9 : FAILURE in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9220b4b615f74b4e980b3ee6d2cdda24 : SUCCESS in 2h 06m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5c980f1655cc4692b53d660d844b6296 : FAILURE in 2h 36m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0048bdb8c3a0422d99362d3952d5de8f : SUCCESS in 1h 49m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5b268eaccecd443285d7f7f69752aafe : SUCCESS in 2h 17m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7379ea80152f4f1f8c0ec95033b2f8da : SUCCESS in 34m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/25364bd56264435d934fae1d0b720052 : SUCCESS in 1h 12m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8b1ec7f63f54196bd1590498ebae1a7 : SUCCESS in 1h 35m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8bf8f16240d141458225fe7b893a34f3 : SUCCESS in 43m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0f1af45512204950ab8eb0f4aefd7b77 : SUCCESS in 1h 37m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0ae975102b2c45338af76d6da2388ef1 : SUCCESS in 2h 02m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/828cc31e4f0a49bf87ddf1ae3b85709d : SUCCESS in 1h 57m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42b303c2612044cdbacf2abca69f7b8f : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":6},{"id":"e5a005caef6387d0668f1cef233a65fd0fd69fbd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-06-13 09:19:32.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_06_2024_09_13_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"f06d1deffb5ca91e57afdcac99f22ca2e6b6f56c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-13 09:22:32.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/464e6738406c4c488ae36b8c55b75e78\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/75513f669dfa43c0b3d35265e2248abb : POST_FAILURE in 20m 33s","accounts_in_message":[],"_revision_number":6},{"id":"ad6e8588fcfd251fa1dc3486164c04b57aa3860f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-13 11:13:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/2d65934/ : SUCCESS in 46m 43s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/6/upstream-check/cinder-tempest-plugin-ontap-nfs/7cf88c3/ : SUCCESS in 1h 58m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"45922778bef30fffd0ac180e1fa16528d75bbd01","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-13 17:14:08.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerStore_iSCSI/48a7991/DellEMC_PowerStore_iSCSI/1429 : SUCCESS in 1h 22m 32s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerStore_FC/a721fc6/DellEMC_PowerStore_FC/1612 : SUCCESS in 1h 05m 58s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerStore_NFS/19eb9bb/DellEMC_PowerStore_NFS/1123 : SUCCESS in 1h 13m 49s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerStore_NVMeTCP/bbfcefa/DellEMC_PowerStore_NVMeTCP/976 : SUCCESS in 2h 05m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"fa6e02f6f52fb835487835756f902969d65e7768","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-13 20:58:39.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerFlex_3.6/816e57d/DellEMC_PowerFlex_3.6/1556 : SUCCESS in 1h 45m 00s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerFlex_4.5/ef01e52/DellEMC_PowerFlex_4.5/451 : SUCCESS in 1h 48m 16s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/f828fdb/DellEMC_PowerFlex_4.x_NVMeTCP/401 : FAILURE in 11m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"8bb0478e0a7911cc377ceca7a56c9963388be6bd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-13 22:40:17.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_Unity_FC/78fd4b5/DellEMC_Unity_FC/2065 : SUCCESS in 1h 28m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/6/check/DellEMC_Unity_iSCSI/d3c6cac/DellEMC_Unity_iSCSI/2088 : SUCCESS in 2h 05m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"42862277b90a0ff838590f3f1c60b609aab2ab81","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-14 04:04:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e11aa31e20f64d9ea96ee69b14a31e14 : SUCCESS in 1h 22m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a548336f3cc04383bb90baef6aeff245 : RETRY_LIMIT in 10m 38s","accounts_in_message":[],"_revision_number":6},{"id":"6b710e7571e1ba76e367e13993e64aad3bf8e006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-14 06:30:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5e3ae34a921c610541cc9fce75d259a617e5e613","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-14 06:32:19.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"3c5cb442d7bb0a7c89147a7d1f3161a16f4a73ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 06:44:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/429e7ec7cbd14d589df6b34b253972f3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b1d8c720ec0449dbcde8d2a0d7b9c34 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5763e94ecfb34fdd897aa29f9c931e55 : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"40d6f57c36bd68ac2355dbadeb9088a33c715eb8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-14 08:32:07.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7bad55b4fa6f4a17a81937e3cfa9a2c7\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0e9acfbfc3aa49db88e946898351cbd2 : POST_FAILURE in 23m 37s","accounts_in_message":[],"_revision_number":7},{"id":"ccc69b5e48bd532ed093816547f05d86a4760163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 08:45:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/162b5eef43cf4139b9fbb9169d5bc17b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9034edc1ceb04f75bc158a3b02e4a789 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4419da87b93f4d0e9e154e5d2f1ec38c : SUCCESS in 11m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d7ec299defb42a7991478fbda36d281 : SUCCESS in 11m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da7c81e074a24ac2a0ed3443a001e486 : SUCCESS in 19m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed83dc0fbe2849a5b7c217758d70d4c4 : SUCCESS in 13m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/2a5920b147654c96ab8a3a34ed74221a : SUCCESS in 1h 02m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c450c055aff4439bc4e85b0859971de : FAILURE in 1h 19m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a74b88ab5f054f33bdaf2485097061a7 : SUCCESS in 1h 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b4dc3c302bd4d9386767033cba23c1b : SUCCESS in 15m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/63476df86a75473faea30f0a5a8e7717 : SUCCESS in 15m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7dc9c5c3e782449c9d5667a24a4737d9 : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/451efbc501f44370993aa6d8f3ce3d4a : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b18b7fb03847472a94da68fa85fae527 : SUCCESS in 6m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68baed3927fc499e9bc531270fbf1a7c : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ba0989e8f46b4346b43d2f378d261c90 : FAILURE in 41m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e9830356a71f434aade4a5f7ad367fbd : FAILURE in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/46707afb782142bf9f7a93b7c33527a6 : SUCCESS in 1h 58m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/df5de5072fe442d792175b78234a1511 : FAILURE in 2h 10m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/227df79145da4e619ec8208250578f5d : SUCCESS in 1h 45m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/61ef8370fdd44a00b0568277484c8d93 : SUCCESS in 1h 50m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/022ecc1b010e42a484097687672effc6 : SUCCESS in 36m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/deefa143519d470ebe6e8c31f5c67459 : SUCCESS in 1h 15m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8a307fa5e92a4446b751d4458a6e32ad : SUCCESS in 1h 28m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/68f9efb8665c43449677046281c28969 : SUCCESS in 41m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dafb12ff559c4c6a882eb1889aa768df : SUCCESS in 1h 23m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6691860a31ff4811a7aa8225b53b1bb5 : SUCCESS in 1h 33m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b28f3385d70746ca960df38a1c8ad295 : SUCCESS in 1h 36m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8ec67fec8b44a2784c603c932d54a55 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":7},{"id":"ad0dcc57fbb9375dbc8dbb69798d3008cef92d81","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-14 09:33:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"48fc25e1e1bfa346d963365b630e1fe6fb610605","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-14 09:35:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"79e98ddccab2db9a0a1ecccce376490592939165","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 09:53:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85feb92f797d467291b5420274c1e4d3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3647fd0b127b4ed6a427ccbf8322e3b4 : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5270ba195e584a05981d3cc7a270cc93 : SUCCESS in 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b3aead5e7054a154f40853033af3c02047015bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:28:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1ece11333ef4adaaefa197486de4ffd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c929e8469d49d6969e0fa4d7c43cfb : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/115c1b21305540ce8c982dda79e5b75d : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59c64824a163452ea1e0aef5938ded9f : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab9b4302600742bc93714d5d9aa651d2 : SUCCESS in 19m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d89c0af8f5004c3db4fa245a9c5fb807 : SUCCESS in 15m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/59cda9b5400e451fa5511b40d8b9dda7 : SUCCESS in 1h 03m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a402f43dedee41ffbf70828074f77e90 : SUCCESS in 1h 49m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed0f5006116749238461f658218cfaf1 : SUCCESS in 1h 03m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b278e7bd8ef3427eaefa38763536192e : SUCCESS in 17m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/942ab69466294ed988730100aa1a549a : SUCCESS in 22m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8dd636544d7b4f878d4de4b54543514e : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/610cb3c1863848c89cd6eae686e49a1f : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c0e8f84c4e104f3fa7a075e54e270445 : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5fd0897d5d204c8397fba1126205a1b6 : SUCCESS in 7m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6fd7a51b1e824cb7adf3438fa5b22702 : FAILURE in 50m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fff1600f8fa0488a818121ebfa36de1a : FAILURE in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f54b477ea8aa4846aced6984bd1486e7 : SUCCESS in 2h 23m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/352626dcc20b4246bb8f24b5cf4acabb : FAILURE in 2h 39m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ad7ae31e85f94a769f28e776de27e09c : SUCCESS in 1h 45m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b65ce4dacf794f1c8f5fbff72c4763d9 : SUCCESS in 1h 43m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a1cae4860c0d4d55a64fca694569510d : SUCCESS in 40m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d2d1a7a1c044baa84fe17b80c1c1c10 : SUCCESS in 1h 13m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a2aafb55556487fa2b905ed85edad7d : SUCCESS in 1h 47m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bf2d0019ed5b4901a1fae76f53be9792 : SUCCESS in 56m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7b2f673236c147e1a98deb056b0246b9 : SUCCESS in 1h 33m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3be11188f3eb4832b548436d81538d18 : SUCCESS in 1h 32m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/079e93985bb94f3795dad59205582ff1 : SUCCESS in 2h 12m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b457e80c78d47c29d06d5a93d7173e9 : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":7},{"id":"2a23a07d5034b2066103d4b7ce7509f8141fc005","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-14 12:38:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/910483 : FAILURE in 31m 57s","accounts_in_message":[],"_revision_number":7},{"id":"cb0cf2ea536dd9ec7eb9e9b0333c85e850bbe292","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-14 12:39:32.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/aaec984466644eb1958dcd659d0fa93c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/50e160b6dce14312b64480475aa8ea0b : POST_FAILURE in 21m 45s","accounts_in_message":[],"_revision_number":7},{"id":"dcacc45b6801fea499823eb570f039ddc065ff78","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-06-14 12:41:04.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_06_2024_12_34_52\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"19050b577f55d475c479cba51f243fd4208b07e8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-14 13:33:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 35m 20s","accounts_in_message":[],"_revision_number":7},{"id":"402cbc6ff7c0ed9665cd862561f41ff7d4ee74d0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-14 13:54:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/8f3d86b/ : SUCCESS in 50m 12s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/7/upstream-check/cinder-tempest-plugin-ontap-nfs/43105b1/ : SUCCESS in 1h 24m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"60a50fd8bb43ee4b0e094e65e6284eabd497f0f9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-14 20:16:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/114/.","accounts_in_message":[],"_revision_number":7},{"id":"9b1625dd9069d471d0ef02921425f797da3390f3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-14 22:29:59.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-iscsi/14d05d6 : SUCCESS in 2h 26m 43s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-fc/448843d : FAILURE in 2h 41m 25s","accounts_in_message":[],"_revision_number":7},{"id":"fd74f17acfdfebe3ffee58b79bfa218ad37905af","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 00:47:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/600/.","accounts_in_message":[],"_revision_number":7},{"id":"a10f8b9126a48dea7e678029263f8a4adc326a41","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-15 01:42:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7c221d0d752b433d84865e3a4f790a8e : SUCCESS in 1h 16m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0146c618b29d4602bbd569a88618ab0d : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":7},{"id":"9a9a387b49de219d1b7b21d3058b5ef975e71f7c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-06-15 06:00:08.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/7/fc : SUCCESS in 0h 34m 29s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/7/iscsi : SUCCESS in 0h 51m 44s","accounts_in_message":[],"_revision_number":7},{"id":"e6c2f690eaad5335a076f166137daffb50c9f2d9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-15 08:27:52.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_PowerFlex_3.6/e89ffc5/DellEMC_PowerFlex_3.6/1574 : SUCCESS in 1h 44m 51s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_PowerFlex_4.5/4223711/DellEMC_PowerFlex_4.5/469 : FAILURE in 1h 45m 47s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/9611623/DellEMC_PowerFlex_4.x_NVMeTCP/419 : FAILURE in 11m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"327cb959aa229bf82e22c0363ede18d24ebe4942","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-15 10:51:38.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_PowerStore_iSCSI/b9260f3/DellEMC_PowerStore_iSCSI/1447 : SUCCESS in 1h 26m 04s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_PowerStore_FC/280634c/DellEMC_PowerStore_FC/1630 : SUCCESS in 1h 02m 32s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_PowerStore_NFS/62e5dfa/DellEMC_PowerStore_NFS/1141 : SUCCESS in 1h 12m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_PowerStore_NVMeTCP/514c374/DellEMC_PowerStore_NVMeTCP/994 : SUCCESS in 2h 04m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"f7733b5822691045ca575188cbaf48597c0e9ac7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-15 11:06:06.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_Unity_FC/972ff5a/DellEMC_Unity_FC/2083 : SUCCESS in 1h 37m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/7/check/DellEMC_Unity_iSCSI/d2cefa1/DellEMC_Unity_iSCSI/2106 : SUCCESS in 2h 29m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"7f89e7e655ae5a3a37090897e682ce8629fa18a1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-06-16 00:01:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/910483-6/HPMSA-iSCSI/ : FAILURE in 3:22:28\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/910483-6/HPMSA-FC/ : SUCCESS in 2:29:55","accounts_in_message":[],"_revision_number":6},{"id":"4973fa05966c7be08e7252b2c5c7dc7b8563c160","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-06-16 06:08:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/910483-6/Seagate-iSCSI/ : SUCCESS in 2:23:41\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/910483-6/Seagate-FC/ : SUCCESS in 2:13:41","accounts_in_message":[],"_revision_number":6},{"id":"9dd0fe049e5f0df033962301b3b6b2cde2821c1b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-06-16 11:21:08.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/910483-6/Lenovo-iSCSI/ : SUCCESS in 2:21:15\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/910483-6/Lenovo-FC/ : SUCCESS in 2:13:02","accounts_in_message":[],"_revision_number":6},{"id":"370c291414ceefc8660ac3fc037f706f76688e6e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-06-19 07:22:27.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/7/20240619082755/NEC-V-FC : SUCCESS in 3h 22m 40s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/83/910483/7/20240619082755/NEC-V-ISCSI : FAILURE in 7h 54m 29s","accounts_in_message":[],"_revision_number":7},{"id":"154b875544d32d5a9872bc2fe3bfc8546d160963","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-25 01:21:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"33d8de42c167f8f2d4ffe5a02d29082b0e11e848","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-25 03:47:17.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-iscsi/c51a93e : SUCCESS in 2h 21m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-fc/8847106 : FAILURE in 2h 17m 24s","accounts_in_message":[],"_revision_number":7},{"id":"ee973c6fc4485aee5981c0a8eba6099f7a75c77c","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-25 05:11:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3a8d763617bbaa7727df2a58e4f1d443e7627e0c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-25 07:27:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-iscsi/ff124ee : FAILURE in 2h 11m 26s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-fc/4cbaba6 : SUCCESS in 2h 12m 19s","accounts_in_message":[],"_revision_number":7},{"id":"9d47a6dcd79073e0d11cce5f35aa5b1e12ff8297","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-06-26 01:35:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/910483-7/HPMSA-iSCSI/ : SUCCESS in 3:15:21\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/910483-7/HPMSA-FC/ : SUCCESS in 2:29:11","accounts_in_message":[],"_revision_number":7},{"id":"f4fc3b90fa1d3cdd44e721b40209169c44ef18a3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-06-26 23:11:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/910483-7/Seagate-iSCSI/ : SUCCESS in 2:18:09\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/910483-7/Seagate-FC/ : SUCCESS in 2:13:12","accounts_in_message":[],"_revision_number":7},{"id":"2e0a4432103873b2aecadce36210ae6a0b646db0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-27 07:25:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"132ce06f93f09a1a145fafe088a57931a04ba599","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-27 09:52:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-iscsi/ebb865b : FAILURE in 2h 23m 42s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-fc/7674415 : FAILURE in 2h 21m 14s","accounts_in_message":[],"_revision_number":7},{"id":"f0f41525a7bf43dfe4946557e1da944a8e9a22cc","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-28 02:20:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"328a38b29164941b77967a1edfbe73197373056b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-28 04:41:10.000000000","message":"Patch Set 7:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-iscsi/1d293e4 : SUCCESS in 2h 17m 11s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/7/check/fujitsu-eternus-dx-fc/f0594c1 : SUCCESS in 2h 12m 35s","accounts_in_message":[],"_revision_number":7},{"id":"80ef7d8680fd5c54cb3e3f52a190a60e5c7cd1ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-29 06:17:48.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"6b0adabe4abb45b2d5e9bacddaa53775eeb2977d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-29 06:17:56.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/433/","accounts_in_message":[],"_revision_number":8},{"id":"2ac63eafe7d69c696cb2f3d6d83e7df0bcb93595","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-29 06:23:03.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/8/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"65ebec368ed47dc5524342bd1dd1a23fce7d1b51","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-29 06:34:19.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 14m 59s","accounts_in_message":[],"_revision_number":8},{"id":"7fc060317ebbd9c972f34ee926d83ce6686e89a8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-29 07:44:54.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d33cc1b431244e13a29e5e2970f80b4b\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b6eeba4198044af0ba13c0c7e639c5b8 : FAILURE in 46m 50s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/560b830db9424470b85e33bff95baa25 : SUCCESS in 40m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e166f31aaec489a888c941ce8c8b9f5 : FAILURE in 1h 26m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b0771b2b70b478fbddb2b16653da60a : SUCCESS in 36m 16s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"cd3525033ec09457cf176bf8bd7efe6a5fc143fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 08:03:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68116c42496a461785b18471797fecf3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0cf987511aa4d0e94eac07df047a364 : FAILURE in 31m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0df68cff7484b078f5046dc013421c1 : FAILURE in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e744279bc03b67141dd7a12b8c19ab00c3083755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 08:50:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa63dffbc6da4b6aa316ac8f75090319\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbeea21a51d34d8db677dd3758bca05b : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8846c21025b444f7a6eb24da462b82d0 : SUCCESS in 10m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f16e919890bc45c7a9b1fc519fed9836 : SUCCESS in 11m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/167b28e1e6bd496d8250ba871d7bcfed : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba1bd8a6c9741d7a4d53330f191b12b : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/1fcfb92b4c874d3ca06844c87bfe44db : SUCCESS in 1h 03m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/43c1f3c0be4e4c49b3e07a6fcc650698 : SUCCESS in 1h 30m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/480a699d8ff24ee8bc636f88e01febf7 : SUCCESS in 58m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16f423fc373249949f5110aef8541c9e : SUCCESS in 18m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6edf678d521147d0be081a4bba4a9fca : TIMED_OUT in 40m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/133146d0f69f49fea887cfa47775bb87 : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8bbe1372a4140a6b86a72deea994447 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f9fccb3898f0418ca2caf94ccb393491 : SUCCESS in 6m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f2b889cb66814d1f829a8719cb884784 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1c3cb81ab6564f13bcdde11bf9a78128 : FAILURE in 39m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4cb0969ba2dc4b98bd5f1d2c3ebfecae : SUCCESS in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd9cf801399e4ed2b0efee278bfefa1d : SUCCESS in 2h 06m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/425878d97de24375801766fca6eb39b8 : FAILURE in 2h 20m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4e906371de1643508752f30696929f05 : SUCCESS in 1h 26m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e62ebd61ac534029ba95a9dd09221f37 : SUCCESS in 2h 23m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b3b675dd93b433c83a004598a22c0cb : SUCCESS in 30m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/15733775c72e48a98cbed83385f34754 : SUCCESS in 1h 12m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/da0c69d01b2d40d4812ff597f430bb58 : SUCCESS in 1h 22m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f775af1bd9de420b9551524269ab33ff : SUCCESS in 47m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a705a8d654904527ad49daf3ba156319 : SUCCESS in 1h 23m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9989cda9987e492ea1c4411702610dac : SUCCESS in 1h 22m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b2067f676b247159c55cb5ce506ab3b : SUCCESS in 1h 34m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43bd266f6c3d4f6fa30ad47d5cf8ca69 : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":8},{"id":"d6a0e9121f4c5905d11ba107289683058a19ed6a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-29 09:53:26.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/9cfb6aa/ : SUCCESS in 55m 36s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/83/910483/8/upstream-check/cinder-tempest-plugin-ontap-nfs/ac0ef28/ : RETRY_LIMIT in 4m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"f0ae3d29bf5c8f397bedf4b07eb4086ff2b663b4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-29 10:11:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 36m 53s","accounts_in_message":[],"_revision_number":8},{"id":"2265268e98cb8d1635568fdd7ad349d086f2f487","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-29 10:47:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/437/.","accounts_in_message":[],"_revision_number":8},{"id":"d805fb02940d1dfb5c8d35dbb99d7b535e859b30","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-29 14:48:40.000000000","message":"Patch Set 8:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/8/check/fujitsu-eternus-dx-iscsi/24e85d6 : SUCCESS in 2h 46m 16s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/8/check/fujitsu-eternus-dx-fc/600c30b : SUCCESS in 2h 46m 57s","accounts_in_message":[],"_revision_number":8},{"id":"7831e2ed5aedbd1a5480ca2feb5da9e4edbfeb8d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-29 15:02:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d0abc3baa83b4011a48fa401a2c916a8 : SUCCESS in 1h 07m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dd640f44fcde4b279e05ddda12f8cc3f : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":8},{"id":"d1d12aeed4edd898bd3828718b6937db95e27986","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-29 15:09:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1026/.","accounts_in_message":[],"_revision_number":8},{"id":"01964a99145219ac71031e587bd323bd8c7cfd68","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-29 22:32:52.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/8/check/DellEMC_PowerStore_iSCSI/f60a8a7/DellEMC_PowerStore_iSCSI/1672 : SUCCESS in 1h 20m 07s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/83/910483/8/check/DellEMC_PowerStore_FC/a29e4a6/DellEMC_PowerStore_FC/1868 : SUCCESS in 1h 05m 31s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/83/910483/8/check/DellEMC_PowerStore_NFS/0b0bba2/DellEMC_PowerStore_NFS/1365 : SUCCESS in 1h 23m 53s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/8/check/DellEMC_PowerStore_NVMeTCP/7354e48/DellEMC_PowerStore_NVMeTCP/1226 : SUCCESS in 2h 01m 43s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/83/910483/8/check/DellEMC_PowerStore_NVMeFC/5eb573d/DellEMC_PowerStore_NVMeFC/189 : SUCCESS in 1h 14m 06s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"705a9fcfcad33c13c9b709f46b3163a8e618bdb7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-30 05:19:21.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/8/check/DellEMC_PowerFlex_3.6/f56ba04/DellEMC_PowerFlex_3.6/1999 : FAILURE in 13m 44s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/83/910483/8/check/DellEMC_PowerFlex_4.x/540a637/DellEMC_PowerFlex_4.x/881 : SUCCESS in 2h 09m 54s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/e07f9ff/DellEMC_PowerFlex_4.x_NVMeTCP/852 : FAILURE in 13m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"baa96cc3914b8119af561d03245a037220e2f6ac","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-05 05:15:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/910483-8/HPMSA-iSCSI/ : SUCCESS in 3:16:12\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/910483-8/HPMSA-FC/ : FAILURE in 42:45.77","accounts_in_message":[],"_revision_number":8},{"id":"ca3a4f298cf68e7b22a9ff7646c12221945847c7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-09-05 12:29:19.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/910483-8/Lenovo-iSCSI/ : SUCCESS in 2:23:40\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/910483-8/Lenovo-FC/ : FAILURE in 2:26:06","accounts_in_message":[],"_revision_number":8},{"id":"f985b09f41f7187be9867c4f76917e6415d01103","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-09-06 04:56:36.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/910483-8/Seagate-iSCSI/ : SUCCESS in 2:23:32\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/910483-8/Seagate-FC/ : FAILURE in 2:33:38","accounts_in_message":[],"_revision_number":8},{"id":"0723a098a03bd8dc4ddf48cf6d90d9689d52d631","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-09 01:32:47.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/8/20240909061416/NEC-V-ISCSI : SUCCESS in 3h 35m 57s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/8/20240909061416/NEC-V-FC : SUCCESS in 4h 18m 27s","accounts_in_message":[],"_revision_number":8},{"id":"2fc2ea9da050a4004b3d70bf1c3dccdc161d46ee","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-12-09 09:34:41.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/8/20241209124519/NEC-ISCSI : SUCCESS in 5h 23m 56s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/8/20241209124519/NEC-FC : SUCCESS in 5h 49m 18s","accounts_in_message":[],"_revision_number":8},{"id":"6624a061b7c6a33674c36fe1cb30b2843edf8673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-12-13 05:59:14.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"380de3d374653d5357a19681779a4c09615fbaa3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-13 05:59:24.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1088/","accounts_in_message":[],"_revision_number":9},{"id":"4031b6ab6282e3d556452f9a8a48a257ddafa800","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-13 06:00:18.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/9/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"5c211d00b5635641ddb49190453dcb78fb7ed49d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 06:19:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d330ec62564179912198d984717aaf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd9ac74ff100480e9272976bac85698b : SUCCESS in 15m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84dec8f17c27440995eb6aebe03876f1 : SUCCESS in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f437268a986921729f8f528eb1cf70dee288d9e7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-12-13 07:06:18.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/910483 : FAILURE in 31m 57s","accounts_in_message":[],"_revision_number":9},{"id":"681a86e9724b87448e655e544e8abac42b75a57e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-12-13 07:10:47.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":9},{"id":"c7dfc84812aeb1ddb818e5f516fc30862694f98c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-12-13 07:16:08.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 36m 03s","accounts_in_message":[],"_revision_number":9},{"id":"33d981a87bfb3ae9ffd824c443004a8111922868","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-13 07:37:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/71c848ec8c06445183c2eddd8251a89c\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/be7f74eeb3dd4c11845224b9ccbe7f37 : SUCCESS in 39m 54s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76478a7224cd45baa26a33ea15899e7b : SUCCESS in 47m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bc16eea9ac1242ff9e6500af3d03b579 : FAILURE in 1h 34m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71bee8fd4b314d309640e6dc92bd06ea : SUCCESS in 43m 41s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c968c8677b4849e5b4561ed8b2baf9eb : RETRY_LIMIT in 36s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/53fce95e222b4f0589f66739c65fc232 : RETRY_LIMIT in 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77996b71f67a4122be037ba43e955c68 : RETRY_LIMIT in 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1bbf793351ba43208f46877b71bbd06b : RETRY_LIMIT in 36s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":9},{"id":"5f873f039663e9a71cabe60e3b8b50ecb1088698","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-13 07:45:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/106/.","accounts_in_message":[],"_revision_number":9},{"id":"83a85e293454c22b739bfe7c80b69c84d9bae0df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 08:17:10.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2329e661ffeb40989198a0a0b43b154b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b452c0e7ca4bccb4557d4e40aa77f8 : FAILURE in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a53e6e5a46d456fba3808dd8a3c45d3 : SUCCESS in 12m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0fe2c6547eb34ecaa0250c23c4c48c4f : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc12c47a42ce4ccf87d33fa9ce685720 : SUCCESS in 14m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/dc6c65e6275845e3a46db9bd0fff05e1 : SUCCESS in 59m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3fe34cdc51a74ab98c091c4d11dfe616 : SUCCESS in 1h 06m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd8e877a3d504c71ae8f65ec44231bc5 : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c62f1cfa6af7400ab33e2eccd95e1a84 : SUCCESS in 1h 15m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58aaafb465f940b9b7da24d8dcb424ef : SUCCESS in 13m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/119f4d0c7e624bd0b8873bf3639958b9 : SUCCESS in 24m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6118b41af97e4e6492b32d0459d0b814 : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f4781d8df730457db1f5c1381587cfbe : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94cf2a8ca94c43dea4d98ffe98ce6f1a : SUCCESS in 6m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/631cb8c2383e4bc6a7ae82ae82ce1374 : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9cca3605df5a437ba5bdbe025039c40b : FAILURE in 22m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff69f4eee3f04621a8136e86c93f16b0 : SUCCESS in 4m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fdffc95ff11b4946aefb4c59bbe3d7f2 : FAILURE in 1h 40m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a9bd3a713d2c4d72af0778d3ce3eb617 : SUCCESS in 2h 04m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/445bd70d777c4c238e161e66c4306ea2 : SUCCESS in 1h 38m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cd9e3b39c8964f20b5a8e8bbc284c746 : SUCCESS in 1h 37m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b6cdfa3e3bb7437f902bc276df4270cb : FAILURE in 46m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d787e031b5bf4d84892eeb6eae68a12d : SUCCESS in 1h 12m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/802d5f7fb4eb4123b8330ccb57e7aa01 : SUCCESS in 1h 38m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/235528f457f1413f8e3c9eb1820df440 : SUCCESS in 52m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/06491dd45fa5447796cab04bb7bfcbe9 : SUCCESS in 1h 31m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f95d965b84b949858754b0c7a237197c : SUCCESS in 54m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/318a17b4eaff4517827e63e209f28e7f : SUCCESS in 1h 44m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46f4726d6da341b48941f6f706d61ebe : SUCCESS in 1h 27m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/464c3817f71045508f5479c9ee369f3d : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":9},{"id":"47e74237b06af5b3138234e578d3d544b96119ce","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-13 08:52:28.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/10b752cfd37344f1b84168a1bb91b3b1 : FAILURE in 1h 24m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0b50bad3ee1342d9af2a0b161ff6ee7d : SUCCESS in 1h 22m 46s","accounts_in_message":[],"_revision_number":9},{"id":"181a22bc9077d220a481d6e3b1fc0e71fbea6983","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-13 09:37:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/105/.","accounts_in_message":[],"_revision_number":9},{"id":"0bd180df87dbe7daefde34e9231e211aebf360a5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-12-13 10:00:09.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/9/fc : SUCCESS in 1h 00m 52s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/9/iscsi : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":9},{"id":"4e7900caf90f33b9c355eb2305a73eb74d3f4975","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-13 21:37:59.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/9/check/DellEMC_PowerFlex_3.6/b35fa90/DellEMC_PowerFlex_3.6/2575 : FAILURE in 2h 47m 12s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/83/910483/9/check/DellEMC_PowerFlex_4.x/38fa8aa/DellEMC_PowerFlex_4.x/1414 : FAILURE in 2h 03m 23s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/78c660f/DellEMC_PowerFlex_4.x_NVMeTCP/1290 : FAILURE in 4h 21m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"84e69556624b0f7f08e2ef34b6cf77f015283bbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-12-16 01:49:00.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"f9a913fce07dad6e2bc8f89ca98348cbcbca2476","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-16 01:49:07.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1094/","accounts_in_message":[],"_revision_number":10},{"id":"c91cf1c6d3379679e52c0dd334e13ab2f9c7c4eb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-16 01:50:01.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"d04bf7b4513d2bba1f5bddf79630aec463c53a04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 02:20:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ce77871ca4b4bb88bbdeeb63c2baf3c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61e9ceb1b01340579eb4b0b64e9e9510 : FAILURE in 28m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd1dd744f89447d583db43bd5a39adf5 : FAILURE in 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f4c14179ce6ab9a153d3bd25bb496b26263c1b55","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-12-16 02:23:55.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/910483 : FAILURE in 32m 12s","accounts_in_message":[],"_revision_number":10},{"id":"3e7b1e8075b736bf5fe90ea75a4c1fa8e77b0bde","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-12-16 02:26:04.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 34m 11s","accounts_in_message":[],"_revision_number":10},{"id":"e053b3ed845d1d13cb9da0581a5e054d850dd928","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-12-16 02:26:14.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 34m 39s","accounts_in_message":[],"_revision_number":10},{"id":"9f85a03c1a56465356aca7d46852b994020394a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-16 02:43:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/118/.","accounts_in_message":[],"_revision_number":10},{"id":"2348e47f9edc2a0a04f422c61baa3f65a8ea9c3d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-16 03:20:45.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/60df2505d0eb46d7aa9f8f4f93fcae43 : SUCCESS in 1h 12m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9010d6f3f07b46859326ff9b0d9c5b67 : SUCCESS in 1h 29m 04s","accounts_in_message":[],"_revision_number":10},{"id":"bed803bc368cdc0188ddfa020e75058128c6ff86","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-16 03:24:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/114/.","accounts_in_message":[],"_revision_number":10},{"id":"37b59b6045770cffcb3e558124f3008f46ed1121","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-16 03:40:53.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8cb572496cad4d4b83938da191be069c\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3cf4317c26604c7ba6c5f1081cb96fff : SUCCESS in 41m 18s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a6ba3de3da2f4da892ba8ae1b2f1e76b : FAILURE in 47m 34s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c369af377acc4c2dbb3cb45a77915503 : FAILURE in 1h 50m 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ebfb7b9e8c794cf585fd702624a6860a : SUCCESS in 40m 19s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d37c422c2b6b44f9b6ea01109c49aeb4 : RETRY_LIMIT in 35s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f8ceba0a02744f93b45494682ebfa89b : RETRY_LIMIT in 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9f599a176bd945329a32049d63213559 : RETRY_LIMIT in 44s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5ecbce297224487ad387831976e241b : RETRY_LIMIT in 42s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"1be6cd6f2d2f7f0f46328006f63dbd712b2f717c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 04:36:37.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9028d929a174de9af6e64e3e4ffebf5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/323f458855964ff7b839d59b12d844a1 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e903f7959af949b8b8f52f52e77957cf : SUCCESS in 10m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d92fb18c32374e1eb29da1696bed4d12 : SUCCESS in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e466e3f2364b83840366a09eca7dba : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/97d78a4b9da24b8cb553ca439cae5df1 : SUCCESS in 51m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/774c1dc70c504739a27247699a747ac7 : SUCCESS in 42m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a952c22713254d3288fd0366053406d1 : SUCCESS in 1h 32m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e0437cb634941629eb96d979345d4cc : SUCCESS in 47m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffd4f779154c4811beb444f13d9a5744 : SUCCESS in 16m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0c64bd75c3a1431d8ce00bee6b358e73 : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da21d485d4f9452a8c47164b8e8cd2f4 : SUCCESS in 6m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/82a4d9e2c18d4a49855d2fbe0bb0b15c : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a96e965b461647beb480125f46e1e551 : SUCCESS in 5m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3c3c7441d3846b89fb1934a95d34e30 : SUCCESS in 13m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/75282bd51eb047cf904c9f0e1c1208d7 : FAILURE in 40m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cfeb1c2733ed4371bf0e4a5b994b3532 : SUCCESS in 7m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7b615b503c84474e9b4c72a63675e331 : SUCCESS in 2h 34m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5fe9adfae62b461c9e8f9d8620a71ac0 : SUCCESS in 2h 01m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6c8a0e21001247e99c064b2d254b4706 : SUCCESS in 2h 23m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9a8f24ff8cb648c79e0ded16d68bce90 : SUCCESS in 2h 02m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fe6333940edd456c83560dc903d4a01c : SUCCESS in 58m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad5967d295fb40c5a6aa4b6d2da48186 : SUCCESS in 1h 14m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b756c35756f4027bfd2518077f04235 : SUCCESS in 1h 56m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7002c22e077c45e18986d17528629c9f : SUCCESS in 1h 12m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2da80eeebcec4570a1769bdb22f59c2a : SUCCESS in 1h 45m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8c891bbed792448aac04cdb64ecfe637 : POST_FAILURE in 2h 41m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7de7891226c419d8b36815885652c09 : SUCCESS in 1h 57m 35s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b01e97145e74ed2864d568f3de02ff2 : TIMED_OUT in 2h 04m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81b15d7e60b24335b990a76ef75ed887 : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":10},{"id":"e414dfc9cd9409381ab5c3844e2cff529be05dc9","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-12-16 06:36:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6e927799b2f64ac2e317b44824b7a32e66b5ad03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 06:56:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edf69bc6c10c43b0a9f3568e6112d57e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c38e8a1325394e6dbba9436bbe37f141 : SUCCESS in 15m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6cdd425cca224cc79cec6a65140d529d : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5df82ae72a4ed0801ae93cbb989b211bcb005f8c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-16 06:56:10.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/83/910483/10/check/DellEMC_PowerFlex_3.6/9356474/DellEMC_PowerFlex_3.6/2581 : FAILURE in 2h 53m 47s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/83/910483/10/check/DellEMC_PowerFlex_4.x/1730de2/DellEMC_PowerFlex_4.x/1420 : FAILURE in 2h 03m 32s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/0701907/DellEMC_PowerFlex_4.x_NVMeTCP/1296 : FAILURE in 4h 34m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"617150c65d4528f21666331820597ee00c69fd04","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-12-16 08:01:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/1fb07da899944b5eaf6fda248654ef54\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/eb196f5c40eb4204a4c3dc2bfc63e377 : FAILURE in 1h 56m 21s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/a9ffa59919cd460f890dc2dcf5631087 : FAILURE in 1h 50m 42s","accounts_in_message":[],"_revision_number":10},{"id":"94277bcde4d922b89a9b54f889b35f754e26db8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 09:17:17.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a49e76b0f614a36b5e44c2dbdf842aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc8f0c9ebb5a47568c63b9876d43b618 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0202e836eef64449b337ca91bded5b6a : SUCCESS in 10m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70d7c26de47e46298a468910a67f63b9 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d526482c6a04a2baab3505a7b189369 : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/58fa344a8d75470aaab6b502d5595e27 : SUCCESS in 1h 02m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11aa95548b8e4b6c8a649a30a12401f2 : SUCCESS in 51m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cb01471efcf4be191e9b87cf8b13c66 : SUCCESS in 1h 29m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c04d2ec660140e4a91ee75bca442c7e : SUCCESS in 1h 00m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d94b26a361ae44a5abe6920b2b5c4896 : SUCCESS in 17m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d30d77d9480a450581815d788ccfe4d6 : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f51aafc84b5445fa99808be107456ed : SUCCESS in 7m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7c5c267e1de4dd196373ac6e198453c : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d39fd40471974449a04e1aa71de3fbe9 : SUCCESS in 4m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4fcc19730ea74eceabd468c9cc81576f : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/545c5e6006b3422e87f0ec76ed63dc7d : FAILURE in 36m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/082138bf16da4f209ec8149ca95ad87a : SUCCESS in 4m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f8d6e12ec974457b3fab789175228a2 : SUCCESS in 1h 19m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cfad318221424c13af34ac3c91f93afe : FAILURE in 2h 31m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf18cf6cfc19432c86e3411fe3b870dc : SUCCESS in 2h 24m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f93d91cf90624b2aba496313b55a5729 : SUCCESS in 2h 00m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ad8e0e12d8694ff1b17f73902a7d1ee1 : SUCCESS in 44m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8f7526b44e5242f7bd423d9cd9b8ebf6 : SUCCESS in 1h 17m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/70bb44d0985a49c888ccb54853d1e769 : SUCCESS in 1h 40m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/81c19ff4c5ab4231a81f7484189de88a : SUCCESS in 54m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/62dac09481844fa0a983f65460fdd143 : POST_FAILURE in 2h 25m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/064cd837cc45457084d18e6b5f6513d7 : POST_FAILURE in 2h 25m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd1a1a1d234b41dcb0b7305217ae9278 : SUCCESS in 1h 48m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca327e6f1d954027b20c28daf4dd188f : SUCCESS in 1h 36m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c6ed22b063543068f03a951b5c27155 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":10},{"id":"f582b07dcc103227706232a5f9f2b694dc8889e3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-16 12:04:35.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/33b5cc6344ec4a49a7069427fcd88443 : SUCCESS in 1h 15m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6c0ae97a23904cbf87f7fb9508a77af5 : SUCCESS in 1h 34m 00s","accounts_in_message":[],"_revision_number":10},{"id":"f7a4a4e7e7a0e2ff93cef4cfab816875b26e51b0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-12-19 00:27:31.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/10/20241219053521/NEC-V-ISCSI : SUCCESS in 3h 40m 14s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/10/20241219053521/NEC-V-FC : SUCCESS in 3h 52m 07s","accounts_in_message":[],"_revision_number":10},{"id":"75a1f3aa557028d7c06ef3ee5ab992f4c437d749","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-12-23 14:17:07.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/10/20241223172813/NEC-ISCSI : SUCCESS in 5h 26m 57s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/10/20241223172813/NEC-FC : SUCCESS in 5h 48m 51s","accounts_in_message":[],"_revision_number":10},{"id":"20fdfc99c86bdb36d28605de86f2a02c0d8e33bb","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-12-23 18:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/10/fc : SUCCESS in 0h 48m 01s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/83/910483/10/iscsi : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":10},{"id":"d5da9b0756e0d20a6b76c624dfa3ea5060e1b76a","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-03 05:48:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d3495047e1a16c799acccc3807bfb6254b3113b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 06:16:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/694b4b8edbd84cfea4e5adf13f6a7ab4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4603b5375614a03a0107ce8fe1b7644 : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a967ef8d4b6241028d7b0cb7a25ad18e : FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dabad67873cfee6ed7e7ffcc1a83ec0bc00393ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 07:58:39.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd050e6f43644e80bd5d0938ddd2c771\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74d3db72e450447da5321e184e4f3730 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ac6334c3ef343d0b9ff41e273d26fdc : SUCCESS in 10m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8cd767b440b49059301a8b785c033a9 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e2c3143c5742ef824c4c7b8f39ad7e : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/0a39fcb8a1d748a898d6e7e90deafb7c : SUCCESS in 1h 24m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18fbeb07380d4566a8c353d1e0d660d9 : SUCCESS in 58m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70282bc00dec4eae9f74cde66a8a3a77 : SUCCESS in 1h 36m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8f0584e0957484390d59787bfd9cde5 : SUCCESS in 1h 02m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c40a059f74942bfae2e55d5c84b9f23 : SUCCESS in 14m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cf6676b02e874910a622320d8c5b1d2f : SUCCESS in 22m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c5fa5e2a8fff4313923799e3d1d3192f : SUCCESS in 6m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2313f83b3e15438d913891baa335256d : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39fb223b2d884ed486cbfd25c963094d : SUCCESS in 4m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/45af19a3eafd46f5b8e62c8547624b79 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b41bd76d451746ffb4767d2ef60cd701 : FAILURE in 19m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f8bb0e5ea5846cfa60f3cff3e59a1a9 : SUCCESS in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d1fd2007c99c423ebdcd394fb67afd8d : SUCCESS in 1h 50m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/108ede49eac14e23b8a274c5795d6d9c : SUCCESS in 1h 58m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/80211c5a80044523a3096973161b300a : SUCCESS in 1h 26m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c4baa70eb513405dab306b840458ced0 : SUCCESS in 1h 37m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/03b60fe88b984a44a806a706e5fae63d : SUCCESS in 39m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fa311469325547438547dcfb17d4a5da : SUCCESS in 52m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e4fd9b7d2ae24e5b8e8a04471c6aca4e : SUCCESS in 1h 35m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/560c0ab577e24d79895f3ff697a168c5 : SUCCESS in 52m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f8d80711e7b14a71bf12b8bd3adba278 : POST_FAILURE in 1h 42m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e7b69377a07e43f7bcff273da9822f00 : SUCCESS in 1h 39m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e341c15bd7bc4c9ab4f18831a1f8d4e7 : SUCCESS in 1h 13m 48s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16bf587c803347f6abb1d81299f66590 : SUCCESS in 1h 30m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e60673655974f3db7bb08e50eee4b8f : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":10},{"id":"0df54461df59735e33319881c3104ce6b7196785","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-03 08:32:32.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6c2f16ad7d6a320be3e530824b17979408c71b19","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-03 08:50:13.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/459042e3e4de41a4b06cb60c9352046e : SUCCESS in 1h 16m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7f89b79bb84a46d3b2945b5e73d3c130 : SUCCESS in 1h 28m 22s","accounts_in_message":[],"_revision_number":10},{"id":"b8e7add78cec9965e848736fcdf5351ab0e0fcf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 09:02:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27356f53932f43ce96cab0292e1e9635\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b1ae2f5ac224e5d97bf3fd7e2410022 : FAILURE in 25m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d832094bc7c4002888a00c27311f1de : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"12dfdbaba35723776dda6849d97a0e4577b2826b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 11:03:56.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d9632b80af34ecabb600b37c5671478\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a1b08f7bf949af88239e8d95dbe795 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f38657adf59b431f870d6b15dbf150a8 : SUCCESS in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09b2a01245a143abbc0ba4700ada7df3 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2019c12c1871460a9d400e6f2b5e1447 : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/c04add00bbe741e78ae253d281d25b31 : SUCCESS in 59m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7c1a965ab89f4d80b1e281f0b302dd86 : SUCCESS in 59m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73490b72ef184befa97752b1b72b1e52 : SUCCESS in 1h 32m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c48946e7ede4e809eb0ea0fd3055a39 : SUCCESS in 1h 01m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/500e47543716425e9bbab49fb0e5a8c6 : SUCCESS in 18m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9e8aa76e311e4f19a9aa858118a6c40b : SUCCESS in 23m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2e46dad10065431dbbb477554e143247 : SUCCESS in 7m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/db29c342672b42a397c26b46fdd913b7 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3d05474c9b0348b4a635a9823252d7b0 : SUCCESS in 6m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ddd05b22dba3431a852eebb0dafa6242 : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dba234ee692149e0b775ef49f190a8f1 : FAILURE in 38m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7ba41fa163a43f7a9cb2c0da156da22 : SUCCESS in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68aec9c97b594271998559dd2e8f5b81 : FAILURE in 1h 48m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dcd833a5c1c14dfbbfdcb385090a5a23 : SUCCESS in 2h 23m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52d85ed2629b464fa77f0a3e8d9fc6fc : SUCCESS in 1h 32m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9b8b00823ac347418dbaf802dc009e66 : SUCCESS in 2h 17m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b59d09b0f32749cbb19dca0326af2b8d : SUCCESS in 43m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f05c3463291c478084b91c27dd637329 : SUCCESS in 59m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e0ec87bd67204941a9b0bfc510751055 : SUCCESS in 1h 17m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/efb5236c78044a5b8e82b8082036eb63 : SUCCESS in 44m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d57e78bb265046cfbfac4e2d95e4bb78 : POST_FAILURE in 1h 39m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/712bc54ccea2489596f044cc1d855e7d : SUCCESS in 1h 13m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8275ad24987a4419aec46d1914c31edc : SUCCESS in 1h 25m 29s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/622915d7a20e4acd87b6858eadcd4e66 : SUCCESS in 1h 32m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3da28d4c0ea45189935f9e63ab04ea5 : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":10},{"id":"a2fc6b1385511afbd7d28c41c8bd3fd7746e4cac","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-06 01:41:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"eb8b9f0e56d4a3f995193e20e3cd55b2f8356690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 02:02:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0469151549834182b337f1998ff06a8e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0acad79e1c7f4fcb86079381afa659b2 : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4d882fca6304079852679396dc068db : SUCCESS in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d13bd4a2443c12246c8e1bbc8358037180489c90","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-06 03:05:19.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/811c0df00249456ea8125127f6940a15 : SUCCESS in 1h 15m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9dcfeff09d994a67b269b7c963df4045 : FAILURE in 1h 20m 58s","accounts_in_message":[],"_revision_number":10},{"id":"a6b4b8efa60233041ef365b6c3a239e4f20a52fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 03:47:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc0c7b830e8b4372be9d549f069c5301\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489d234271984324a8e46a35016e9410 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/563c3b15197d40cda30028175607ae94 : SUCCESS in 9m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00f4d13eb75444b684853dc820ac4396 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21b588ddf9642afaa6e94b3f65748a6 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/9e7b2b229c6b498ab2dc20e7df6599e8 : SUCCESS in 48m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d83e27694b5548ff823eabd86c3f966f : SUCCESS in 43m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/97b1a71fe8f74b408f204d29d25209b0 : SUCCESS in 1h 23m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6efac672029c4c6d975371aa492b5f57 : SUCCESS in 49m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1a02ba179d445d3b0ce0c426842cd20 : SUCCESS in 11m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f907b70900094a3a9cc5c5dae119f10f : SUCCESS in 23m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/58e4531408c14e8797870eaf3cbd1484 : SUCCESS in 3m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e51b628549a740358fd6f679616a7505 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bcbda01329fc4209ac1373bbc8d70510 : SUCCESS in 5m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2d9509f92392471aad134a9a6c4c225c : SUCCESS in 5m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae202d4eb9c448f68acf4a5d7fc6c3f0 : FAILURE in 39m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f07bd8358e542969ac7c2ab0c62fca2 : SUCCESS in 4m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ac8322c50ac48698d471c258b35cc5a : SUCCESS in 1h 58m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1227e485f488403baf291475b788e8f8 : SUCCESS in 1h 41m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e72263a677d404d8ed38be5620716af : SUCCESS in 1h 47m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ea1032a84245462b8db9a13a9880b574 : SUCCESS in 1h 57m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4d9accb6a59c4f7581662da7ca4cef51 : SUCCESS in 42m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1d9488d539cf4cb091760b7bf576dadc : SUCCESS in 1h 15m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/43f174242ef34a6c84484df1ba44f83c : SUCCESS in 1h 44m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/58f5c3071bb146ea9501e38fcc143ae2 : SUCCESS in 51m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db5a5b3f2bd74d42932d49682e9931da : SUCCESS in 1h 40m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3757a62affcc4a1fbe51ebdd894e166b : SUCCESS in 1h 00m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f9348b1bbdd4263968f300d1db733fe : SUCCESS in 1h 35m 52s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeb27447535a4b5ba2a7a4f5612e79bc : SUCCESS in 1h 44m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7fbc30738cf4e0e95f9b82501f56807 : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":10},{"id":"99ac689dc7cebb2b98b50eb7b6a5c390d15b58c8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2025-01-06 06:47:56.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/83/910483/10/check/DellEMC_PowerStore_iSCSI/96b8cf0/DellEMC_PowerStore_iSCSI/1829 : SUCCESS in 1h 20m 54s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/83/910483/10/check/DellEMC_PowerStore_FC/772580c/DellEMC_PowerStore_FC/2025 : FAILURE in 2h 51m 13s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/83/910483/10/check/DellEMC_PowerStore_NFS/8c7534e/DellEMC_PowerStore_NFS/1521 : SUCCESS in 1h 22m 13s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/83/910483/10/check/DellEMC_PowerStore_NVMeTCP/b08a535/DellEMC_PowerStore_NVMeTCP/1380 : ABORTED in 3h 00m 04s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/83/910483/10/check/DellEMC_PowerStore_NVMeFC/6807e04/DellEMC_PowerStore_NVMeFC/345 : FAILURE in 21m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"e8e7605171982deedfe45c1311e3e857ece5a421","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-08 01:40:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"60d22b41514ba56772f893df2bcb70b1d959af6e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-08 13:31:46.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/10/check/fujitsu-eternus-dx-fc/2dff4b4 : SUCCESS in 2h 43m 09s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/10/check/fujitsu-eternus-dx-iscsi/c06bc97 : TIMED_OUT in 3h 34m 19s","accounts_in_message":[],"_revision_number":10},{"id":"f6d327f3a3ff4f0d04de2a7cac42e9c86d3877cc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-09 18:04:06.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-10/FC/ : FAILURE in 8h 22m 39s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-10/iSCSI/ : UNSTABLE in 4h 39m 40s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"fe8a55d63e9490fced977494fda5f1b8ecd0921e","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-21 03:30:11.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d43aa929970c566af8ef1324ec97aa8db7264890","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-21 07:04:15.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/10/check/fujitsu-eternus-dx-fc/386af7a : FAILURE in 15m 13s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/10/check/fujitsu-eternus-dx-iscsi/d96dfc4 : FAILURE in 24m 01s","accounts_in_message":[],"_revision_number":10},{"id":"e9b60d05640aff1cfe4faff58f16fa29782abf13","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-22 02:19:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cfa4827097da4bbf2b31de55cbd2833d51d8bfac","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-22 05:20:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/10/check/fujitsu-eternus-dx-fc/29fc2c4 : FAILURE in 2h 53m 55s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/10/check/fujitsu-eternus-dx-iscsi/d54bb1f : FAILURE in 2h 58m 38s","accounts_in_message":[],"_revision_number":10},{"id":"df98bf1657c9dd90d6b9cca0050169499c696e3e","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-22 06:54:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2724aeec7494c83deb9c1fdcc00c7cbb1f6ec010","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-22 09:52:51.000000000","message":"Patch Set 10:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/10/check/fujitsu-eternus-dx-fc/21491af : SUCCESS in 2h 40m 28s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/10/check/fujitsu-eternus-dx-iscsi/bc3ead5 : SUCCESS in 2h 55m 58s","accounts_in_message":[],"_revision_number":10},{"id":"a915a68971e0dd45ed3b91f69ddfba40a16a230a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-27 13:38:49.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"5143a43d06bf1c620d45b7cdd4d376bdde62b375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-21 07:06:28.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"9d39eebb9c3a05dfe4f4d9abba178a8f51fb3096","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:06:41.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1541/","accounts_in_message":[],"_revision_number":11},{"id":"90d66af3a88f458316107e2491f4f6c1174372fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:06:43.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"a287ee1455f247918e46bb4deb0f3b2d734ea920","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-21 07:08:05.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"67b6b78816b5df0e31b51126a99cd1c86d34f33e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 07:22:11.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/207b41917d1249bfb9f25c338ebf944c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a686177bbb6449bba3cd59e443947c3f : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7575c12d364485691ba92541a1de1f6 : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5dc8bf0f5dd8f22b227e625ce6009910361657e6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 07:48:38.000000000","message":"Patch Set 11:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/478/.","accounts_in_message":[],"_revision_number":11},{"id":"a50636e2e574f3922a816f88d77d288a84de6089","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-21 09:34:26.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c91b319fc56247f092bcab4669ec0d37\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/388ad3047a1c43979d112c7f1b247384 : SUCCESS in 41m 34s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18804e07aadb43d6b772062da6df70ee : NODE_FAILURE Node request 200-0000033561 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd88ee867dde4bbaa671eefd39c645de : FAILURE in 1h 36m 05s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b1b258f9f97a44f29d205fa33ebd5629 : SUCCESS in 44m 20s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7128037c88a14708b27aaa0a294f837f : RETRY_LIMIT in 47s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/07818f4a280a4d018bd70160065dc036 : NODE_FAILURE Node request 200-0000033700 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8626519cf3a0430bbcd796d5d97cd7d2 : RETRY_LIMIT in 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/882f57304175444c96a607f98b2f7c90 : RETRY_LIMIT in 31s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n           ... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"ab3e0c022e884059c1cabfbbd9dbf4334a069b15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 09:52:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/892ef36b124a4b07a9b577cc5938a901\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c45e57f39f46aea266926790dd44da : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f22505e784f4815b606fbfc8b1b7535 : SUCCESS in 12m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76cbc744a79949afb17de34f9de5273a : SUCCESS in 12m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b019775e9be4bcbb4096c65d7c0fe8d : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/ebdf167eb1e7469eb8a98b7ac6aa406e : SUCCESS in 1h 01m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e3fa42df53b84040a69f3ac675b29a68 : SUCCESS in 55m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a83d7d97e79b4c92b3cdb3dec3203289 : SUCCESS in 1h 29m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b18ec210767044b096d7961cb5d0108a : SUCCESS in 53m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24f2fd5f05d340418acf5579b51de763 : SUCCESS in 17m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0c314fb397dc4c48b0e76f14d1f7caf5 : SUCCESS in 26m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4b4e2f362bb84e56b4b95f99a02cc1bf : SUCCESS in 3m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98ad467cf8a64643bba77e37ea61c9cf : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/838d33eba3c641a8af64a4b4ec321f3b : SUCCESS in 9m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2e5591e7b28b4ca5a844776d09faf5e4 : SUCCESS in 4m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e1090017380a4b2380c38bd29f17e225 : FAILURE in 17m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a669d8248b2041a18834e9e09ccdd8f1 : SUCCESS in 7m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c4242ce5e43948149bd7a0ca1dd09f9f : SUCCESS in 1h 14m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0af74eebe3be4f31a81f2f32a3a0b91e : FAILURE in 2h 41m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7221d400e66142f88f023b08a6bd9d05 : SUCCESS in 1h 45m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3580f9cf83224cf3ba78591bd4274bf0 : SUCCESS in 2h 05m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c826a5fb7d0a4f50ba7045bd60952df9 : SUCCESS in 20m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/527d43f609244b3180fe07d0008a6377 : SUCCESS in 34m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b3bb35ae778645048c040da2b0915dfa : SUCCESS in 1h 58m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1775bc7a89754935b3bafde820fffaf3 : SUCCESS in 38m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/02b13712be004b7abc8143c8803bd31e : SUCCESS in 1h 55m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5c306f246f4e42f1b90c8081b720a0fe : SUCCESS in 1h 38m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da874bb823624887bc4cbedc78aed076 : SUCCESS in 1h 47m 43s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27a086ecb51042ceb8b0f6433d2c4cd7 : SUCCESS in 1h 32m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8388f8a353374a738115fd327b2ea325 : SUCCESS in 55m 29s","accounts_in_message":[],"_revision_number":11},{"id":"957eaa942cb8fd2f24d58d37ffbc13c1a81f0f9b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-21 10:16:10.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 33m 16s","accounts_in_message":[],"_revision_number":11},{"id":"a7291b8c306f5b5ec0d02c5c8445d73032e62d99","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-21 10:16:13.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"fbd302426cf39637d8c958e8a7eed38c09589b93","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-21 10:20:46.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 34m 44s","accounts_in_message":[],"_revision_number":11},{"id":"c96dbf8fdbb056832eea9c98b0ed1885ebccfbaf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-21 13:58:06.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-11/FC/ : FAILURE in 6h 51m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-11/iSCSI/ : UNSTABLE in 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"d2128f7761e66ff025175baa83ce60f36d79c09d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 14:49:13.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/547/.","accounts_in_message":[],"_revision_number":11},{"id":"0550174fa8c32a73970fc0d92143939b02d9689a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-21 15:50:24.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5b43c49f8b7b45469ef7276a8361d834 : SUCCESS in 1h 13m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/15d4764454b34391983da1806c86a7b8 : SUCCESS in 1h 22m 28s","accounts_in_message":[],"_revision_number":11},{"id":"2d628721cf45b744c9856e8cc7279600b8670090","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-21 17:53:07.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"902a4008c6c86b5e62aa1f8e66fe3ba53225e560","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-03-21 23:18:41.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/83/910483/11/check/fujitsu-eternus-dx-fc/c4c4dab : FAILURE in 2h 33m 20s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/83/910483/11/check/fujitsu-eternus-dx-iscsi/0c41e35 : FAILURE in 2h 43m 42s","accounts_in_message":[],"_revision_number":11},{"id":"33f483a8b42e0d7cb84602713c474845809cd877","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-23 18:08:18.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/11/20250323182607/NEC-V-FC : FAILURE in 7h 34m 56s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/11/20250323182607/NEC-V-ISCSI : FAILURE in 8h 42m 08s","accounts_in_message":[],"_revision_number":11},{"id":"e5eefdc703ac4473344b7f1b6691b447a19430c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-26 02:51:28.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"beb16e3d5426ca69edaecebc297c755ff147abc5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 02:51:39.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1572/","accounts_in_message":[],"_revision_number":12},{"id":"d998f4294a8ee1bb5f274969a3338d9239a40a32","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 02:51:42.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/910483/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"7a75e113f38abbe39ef4eda1f5ca62e100404cf5","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 02:53:04.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"d749e769d19d13b6b2ef7c198099c4cc4c11bb38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 03:23:38.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c33eb88d63f0458dae2b65e654e089e9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23a6d0a17b614c6380da9b4c8f1010c7 : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/306ef3b049d143269bc30aa5a23d918b : SUCCESS in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"94c6adcc630b30dfa34d33fd51c2eded8e59ec04","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-26 05:09:55.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-12/FC/ : UNSTABLE in 33s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-910483-12/iSCSI/ : FAILURE in 2h 18m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"9221032cd3458e14fac5135b9144fd5cb9a53f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 06:26:33.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2f5bed8f8584331ae0e5609791199ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9d2206a6724a64b0ce545ef635bbef : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3de404228e344afd993c2c49b72f56c6 : SUCCESS in 13m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a0abf8ab6974d90891d6ed0eeb0beb6 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c029b370e1147c398d5ca24d4f1af7f : SUCCESS in 19m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/7e148e408b5e49c5b2040beac4e3b2c9 : SUCCESS in 1h 05m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d242d1209d440dfab1fe696184d5396 : SUCCESS in 1h 03m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b51819783b5343d498bbc9d99719b109 : SUCCESS in 1h 30m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5a6d4910e1d48429979f550aaf4689e : SUCCESS in 1h 03m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/992b2b1d050f42ccaabb4ab5958b031e : SUCCESS in 21m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c2a39d4e56ae43acae71039f6bba419e : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0ec5b9a0155f423f8a676c6ac2e2c85f : SUCCESS in 6m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7755d609968e4d689fb27bbf26cdb8ca : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb73d7926c944f13b7bb0ca2970987b5 : SUCCESS in 7m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2d672c891fca4012b49a5f727547ad35 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/260cb884c8444d35a1903417494093fa : FAILURE in 34m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10ee4b0b14334a219b71dcc2c63ccdaa : SUCCESS in 15m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9bc7ca9706fa4352b87ef53699b4352f : SUCCESS in 1h 56m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d24e96e980546ee9fd17e8f450f5d4f : FAILURE in 2h 51m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13f3d47a2c374d6db9dcf91c6dd8c895 : SUCCESS in 1h 42m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1ac51ba4711c4704885d912cc008beff : SUCCESS in 2h 00m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2b61ef9a110649458a2b8b67e4c2da2a : SUCCESS in 17m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/018a6c1f694b4ef5983577010f838e45 : FAILURE in 50m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f286c0d1e85e4a90aaf11be997972949 : SUCCESS in 1h 17m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d3a28692c74349808534bcd289614f8f : SUCCESS in 36m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b78d3fc61df844eb911a014ea4b1665b : SUCCESS in 1h 22m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/002ef5476e7c4bdda17ddc4483c821c2 : SUCCESS in 1h 36m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6a23ffce3344cfb96278d8af52ae3fe : SUCCESS in 1h 30m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c55e6ae36ac64a0b869f0966a24ae87d : SUCCESS in 1h 17m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/575c1b09e64e4887b9da13ff3713f899 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":12},{"id":"e3129be1774c6fade7aa5a06c47854530e4bec70","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-26 07:01:33.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0f94f7ccd9274a9db420e7c7eda041c8\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/171790b35e8643378d4a3f20032ea843 : SUCCESS in 41m 10s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d7be624ebb49409aa87e382d5bdff972 : SUCCESS in 43m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/62d9ac612446428d829789ba7c6e450f : FAILURE in 2h 17m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4bd7a7809e3c40b7bce9c7f292194cf8 : SUCCESS in 48m 49s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d6e7867a9106417e81d6338ea337e7b8 : RETRY_LIMIT in 46s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9bb7afc794cc4ad49d086f2eaec50a2e : NODE_FAILURE Node request 200-0000035345 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dff764c6c87041c5b6b3e9d7aa80bbcb : RETRY_LIMIT in 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/368f27d288654c138efbbbc58bce1a44 : RETRY_LIMIT in 40s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"193cb1eb60bf4f6dc7c3ca240b9d7044466abb04","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 07:49:29.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/910483 : FAILURE in 32m 02s","accounts_in_message":[],"_revision_number":12},{"id":"d65439cb9695b0916e6dcf579773edec7ec1961e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 08:07:31.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910483 : FAILURE in 35m 10s","accounts_in_message":[],"_revision_number":12},{"id":"5fdf00c4839e0c86adc946da5063f60a33a89ebe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-26 09:23:05.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"4e95bcbee56fc2025d8f189704980d4f5893d82e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-26 12:42:27.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/575/.","accounts_in_message":[],"_revision_number":12},{"id":"4b05706ca6accedd22e036667b66d20d67996a6a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-26 13:12:36.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/120bc5d689704aa1844b81af35ba3384 : SUCCESS in 1h 10m 02s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/edc99604d64c48cb8d1bff331d46dcde : SUCCESS in 1h 31m 02s","accounts_in_message":[],"_revision_number":12},{"id":"f738146c228f4cea53303448bfe6edf223a60293","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-30 02:14:45.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/12/20250330022726/NEC-V-ISCSI : SUCCESS in 3h 50m 11s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/12/20250330022726/NEC-V-FC : FAILURE in 8h 47m 16s","accounts_in_message":[],"_revision_number":12},{"id":"343605c0795b0cfb71501258d864edb9ad3f25e2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-30 18:41:41.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/12/20250330214700/NEC-ISCSI : SUCCESS in 5h 30m 22s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/910483/12/20250330214700/NEC-FC : SUCCESS in 5h 54m 37s","accounts_in_message":[],"_revision_number":12},{"id":"ee0feb7757f918350f9d8a448e01fd27cda7d932","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-31 01:13:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e3d97431ef1eb92c9b630e0bb9f72479fe417237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 01:30:50.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13c477a78b964992a97e9c694208b492\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c9e4cb53dc94352bf4e87d90c4a50c8 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8459c5e194464f7fa4a47235683ac6a3 : SUCCESS in 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f496bb392529ad356f8c941e8101c48e35b28dbe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-31 02:38:27.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/656f567365624a98ac75814da1568fdc : SUCCESS in 1h 18m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0e501d0a287a46a09c49dfce2f4d7353 : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":12},{"id":"2e3be87c629ee9d407889febdf090ada5b9478a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 03:26:14.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5797df6dcbb046a39b357539de9fd7cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd13fbc64ed469e8988948443b26ac2 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbecd0b7d58a41ae9ececed12059c86d : SUCCESS in 7m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ed4e87cfdbc41c181edd1a1108391b0 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9983a7a91c64ee39446ef63fc27ef43 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/1ec8a2f92f9e42ebb16286c19144323d : SUCCESS in 27m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67ab28f97eae4bcfb0185880253b210e : SUCCESS in 33m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9847c28dc0f24961a739d1e56522bb5c : SUCCESS in 1h 18m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bcdd03e9fd54f96b7aa1c9747ffc83f : SUCCESS in 36m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bdfec9b42464c45909764d013258f91 : SUCCESS in 13m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a6b023cc6dcb4bc7ab57ebc07cafdc2e : SUCCESS in 22m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ee5b9d32d397436f86d995cb7a0dab7f : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6337f74828ff404f80ca59aecfba38eb : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36ce8f8d4697437bab5aeba802e42633 : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e1e641ec5be7406595583bc53d265c95 : SUCCESS in 5m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed6e372e54fa406d82c0ba1b9c1bca2d : FAILURE in 29m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35b226b5109c4f168882b309d1553b90 : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a8324aa3edad4a0c95df64b9a36d26c4 : SUCCESS in 1h 33m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04a85bae25114133a11d26b05ea7c38e : SUCCESS in 2h 09m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4b80f8fdfb9c44b4b61d495a39d4cca8 : SUCCESS in 1h 40m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/494c985346184cbf95e9009f027556d4 : SUCCESS in 1h 50m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2ef9db5140db483890192878cc6f54b9 : SUCCESS in 26m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bbdd7cd7ac5942e0aefb0fbecbc86f90 : SUCCESS in 57m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3a2489c09f9445f69e8c55d48afa3ce6 : SUCCESS in 1h 28m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/94b43a5cd2394e0ca509e8bd0b3cda49 : SUCCESS in 26m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b46fd7096c4542e485159c5e7da7e254 : SUCCESS in 1h 25m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a70860d437624b238facc780608b7f0f : SUCCESS in 1h 11m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1d811a080a745f09ca2e692c4305be6 : SUCCESS in 1h 18m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edd5252bac1247999a9dac31b0e014fa : SUCCESS in 1h 29m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0880c1613fba4dad9026ec3b5f3e2182 : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":12},{"id":"c6b4a05e3c656439b2d9b705fc425f3ed7a1d38b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-31 06:33:18.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"59024c51d4338a702e5e8ed3ef99f12fe5d9a22d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-03-31 07:52:15.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/910483-12/PVME-iSCSI/ : SUCCESS in 1:56:57\n- PVME-FC https://os-logs.seagate.com/dell-ci/910483-12/PVME-FC/ : SUCCESS in 2:07:45","accounts_in_message":[],"_revision_number":12},{"id":"7e51c418714e5f196e1c25db7c0107b2ab2bebc7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-03-31 13:23:22.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/910483-12/HPMSA-iSCSI/ : FAILURE in 3:03:00\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/910483-12/HPMSA-FC/ : SUCCESS in 2:07:00","accounts_in_message":[],"_revision_number":12},{"id":"3baf49a2ab88ccf667160db86569226c858cd86a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-03-31 17:47:02.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/910483-12/Lenovo-iSCSI/ : SUCCESS in 1:56:07\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/910483-12/Lenovo-FC/ : SUCCESS in 2:06:51","accounts_in_message":[],"_revision_number":12},{"id":"7244ac221b420e062b9cffc6b969dc787a655ce4","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-03-31 22:11:53.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/910483-12/Seagate-iSCSI/ : SUCCESS in 1:57:02\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/910483-12/Seagate-FC/ : SUCCESS in 2:07:03","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"871a392d3529277c834d0154446d564dadb91465","revisions":{"05aa1151b934105682b2df76381d10e11d128fa8":{"kind":"REWORK","_number":1,"created":"2024-02-28 03:50:42.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/1"}}},"commit":{"parents":[{"commit":"307940bb600d9ddec42741d99f9cde17965f7048","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/307940bb600d9ddec42741d99f9cde17965f7048"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05aa1151b934105682b2df76381d10e11d128fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05aa1151b934105682b2df76381d10e11d128fa8"}]},"branch":"refs/heads/master"},"e8f185b61f70b85adc53715bc417e6c1b766beee":{"kind":"NO_CHANGE","_number":2,"created":"2024-03-19 06:23:24.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/2"}}},"commit":{"parents":[{"commit":"c0c76dc7a0463a80f373456cb93d95483c4ab4cd","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0c76dc7a0463a80f373456cb93d95483c4ab4cd"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-19 06:24:11.000000000","tz":-240},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8f185b61f70b85adc53715bc417e6c1b766beee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8f185b61f70b85adc53715bc417e6c1b766beee"}]},"branch":"refs/heads/master"},"d7a2c9eae7b3e4d7b4292b965826e0d365c60354":{"kind":"REWORK","_number":3,"created":"2024-05-16 02:49:21.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/3"}}},"commit":{"parents":[{"commit":"7bc5eab21ea98ff2e461244c83622de198c88d12","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7bc5eab21ea98ff2e461244c83622de198c88d12"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-16 02:45:04.000000000","tz":480},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7a2c9eae7b3e4d7b4292b965826e0d365c60354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7a2c9eae7b3e4d7b4292b965826e0d365c60354"}]},"branch":"refs/heads/master"},"c0975b555a7bf9f521fbe4622f4a6fc5ac476884":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-17 07:06:22.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/4"}}},"commit":{"parents":[{"commit":"de3348d6c5a14ab0de83cef2a1405c5e87f91932","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de3348d6c5a14ab0de83cef2a1405c5e87f91932"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-17 07:05:07.000000000","tz":480},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0975b555a7bf9f521fbe4622f4a6fc5ac476884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0975b555a7bf9f521fbe4622f4a6fc5ac476884"}]},"branch":"refs/heads/master"},"70c609bd82ac045f29ec64a45bdedb0094c11b07":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-24 09:55:46.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/5"}}},"commit":{"parents":[{"commit":"be711c89a849541a6bd22166e67d09934df29b7c","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be711c89a849541a6bd22166e67d09934df29b7c"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-24 09:55:21.000000000","tz":480},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70c609bd82ac045f29ec64a45bdedb0094c11b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70c609bd82ac045f29ec64a45bdedb0094c11b07"}]},"branch":"refs/heads/master"},"3850287e3e6a65390f8640e6c26661dc44882f38":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-27 08:52:59.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/6"}}},"commit":{"parents":[{"commit":"106f15d2212b2daf46fb124bb33306ee2ecbcb2d","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/106f15d2212b2daf46fb124bb33306ee2ecbcb2d"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-27 08:52:33.000000000","tz":480},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3850287e3e6a65390f8640e6c26661dc44882f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3850287e3e6a65390f8640e6c26661dc44882f38"}]},"branch":"refs/heads/master"},"2d244bfa1d4b19af82053a3e4751c27a3e8f2f60":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-06-14 06:30:36.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/7"}}},"commit":{"parents":[{"commit":"88ae53f734e1f739f48344bb6f744f97cfcaff58","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88ae53f734e1f739f48344bb6f744f97cfcaff58"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-06-14 06:27:43.000000000","tz":480},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d244bfa1d4b19af82053a3e4751c27a3e8f2f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d244bfa1d4b19af82053a3e4751c27a3e8f2f60"}]},"branch":"refs/heads/master"},"4c1e22e9cafce0917b270e7723f32f00a947c098":{"kind":"REWORK","_number":8,"created":"2024-08-29 06:17:48.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/8"}}},"commit":{"parents":[{"commit":"10712677a57bc8911b51e852063261fcec902bbb","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10712677a57bc8911b51e852063261fcec902bbb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-08-29 06:16:36.000000000","tz":480},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c1e22e9cafce0917b270e7723f32f00a947c098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c1e22e9cafce0917b270e7723f32f00a947c098"}]},"branch":"refs/heads/master"},"6a28058164bf7b0a29c8a14941e5c7148ab12fb4":{"kind":"REWORK","_number":9,"created":"2024-12-13 05:59:14.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/9"}}},"commit":{"parents":[{"commit":"3233249318337d436c621b81d38b8b1ddc9a2b5f","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3233249318337d436c621b81d38b8b1ddc9a2b5f"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-12-13 05:58:37.000000000","tz":480},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a28058164bf7b0a29c8a14941e5c7148ab12fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a28058164bf7b0a29c8a14941e5c7148ab12fb4"}]},"branch":"refs/heads/master"},"6d9a131e9de88b2142956c92d7251828439717d1":{"kind":"REWORK","_number":10,"created":"2024-12-16 01:49:00.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/83/910483/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/10"}}},"commit":{"parents":[{"commit":"3233249318337d436c621b81d38b8b1ddc9a2b5f","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3233249318337d436c621b81d38b8b1ddc9a2b5f"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-12-16 01:48:36.000000000","tz":480},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d9a131e9de88b2142956c92d7251828439717d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d9a131e9de88b2142956c92d7251828439717d1"}]},"branch":"refs/heads/master"},"d133239dbff17769e8df284dda2671fc9a646f14":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-03-21 07:06:28.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/83/910483/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/11"}}},"commit":{"parents":[{"commit":"1fb4cd2291239c38eca10fc04f8028392582a47e","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1fb4cd2291239c38eca10fc04f8028392582a47e"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-21 07:04:21.000000000","tz":-240},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d133239dbff17769e8df284dda2671fc9a646f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d133239dbff17769e8df284dda2671fc9a646f14"}]},"branch":"refs/heads/master"},"871a392d3529277c834d0154446d564dadb91465":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-03-26 02:51:28.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/83/910483/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/910483/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/910483/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/910483/12"}}},"commit":{"parents":[{"commit":"f2ef6af53a8c8b478d7f82e3543ec1a39d22affc","subject":"Fujitsu Driver: Improve process about copysession","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2ef6af53a8c8b478d7f82e3543ec1a39d22affc"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-02-28 03:51:34.000000000","tz":-300},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-26 02:51:09.000000000","tz":-240},"subject":"Fujitsu Driver: Added checking for cascade copy.","message":"Fujitsu Driver: Added checking for cascade copy.\n\nWhen the copy source is a volume or a snapshot which is set as the copy\ndestination of a 2-stage cascade copy, the wait time must be specified\nbefore starting the copy process.\n\nFollowing options added to configuration:\n\n* fujitsu_copy_wait_time\n\nChange-Id: I5380caaac22ffd50a94aebdc56af47bccc5004e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/871a392d3529277c834d0154446d564dadb91465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/871a392d3529277c834d0154446d564dadb91465"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
