)]}'
{"id":"openstack%2Fcinder~910815","triplet_id":"openstack%2Fcinder~stable%2F2023.1~Ia3bca0e3d6dfcb29087cd680207319e5a663eb75","project":"openstack/cinder","branch":"stable/2023.1","topic":"bug-1895035-antelope","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-06-07 17:32:05.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-06-07 17:32:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3bca0e3d6dfcb29087cd680207319e5a663eb75","subject":"Ceph: Fix restore backups to diff backend","status":"MERGED","created":"2024-03-01 16:57:36.000000000","updated":"2024-06-07 17:34:39.000000000","submitted":"2024-06-07 17:32:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910815-bug-1895035-antelope","meta_rev_id":"9931646b747b263a57e4f589639255ba64c1099b","_number":910815,"virtual_id_number":910815,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-07 17:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-06-02 20:28:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-06 14:31:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2024-06-07 15:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2024-06-07 15:41:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-01 16:57:36.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2024-03-01 17:11:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-01 17:37:06.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-01 18:57:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-02 03:48:49.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-02 08:33:35.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-05 09:48:03.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-06 14:31:01.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-03-06 19:22:09.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-17 12:13:31.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-06-02 20:28:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-06-07 15:41:52.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"0a5f9da9170638191c0ff59bd4ab417f77cfcbac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-01 16:57:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0b923466fb7bc4a0df288f4ad18eb8cef500342","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 17:11:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b924f5448f664f16af76182278cab17f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a76396dbeede444c9ac4d399e48b0687 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58a227f73a104d72ac06198808367cba : SUCCESS in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06f0c8f1b7fb2586e99ff0df85e1402cc6ea6ee5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-01 17:37:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/910815 : FAILURE in 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"06b49b184d54e3a9c6d959ac55f9020949b83621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 18:57:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d8e4adde7b5491ea7be1ade5caa7fd2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/564e401f9e3d4b5ca8bc14db63cf0d82 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be4d63f314bf465090885cc993fc8976 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d67fa918d864855afce61dc0724c1c2 : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/a10f390ba38140b2a8b3639346a2178d : SUCCESS in 58m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/417c2dc7aabc4068af6fca421c9932fd : FAILURE in 44m 13s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d312066faddc4f178d9491081d0fda39 : SUCCESS in 1h 18m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88ecaa46f15a4e10b65eda55d4c79736 : SUCCESS in 1h 01m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13e4a29c7eb94511ab9c16f73d5245fc : SUCCESS in 12m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/db212be0c25f43d2b59a04e349615773 : SUCCESS in 14m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/00e43e236b4b4a4dbba781622ab639dd : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f018fac7fc4c4ed1b41ae09fa50e08f8 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a4033b6321b483dae13744ed8ea2a26 : SUCCESS in 6m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/52d7fa72b1d542fb9d22052e6fff7106 : SUCCESS in 5m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/71de545cb19e451db293b62c4900cf12 : SUCCESS in 38m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7e2ff60483d4001bfa373bf34bae32c : FAILURE in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3dd4ba1172684b58b29ebc2edd4e697a : SUCCESS in 1h 43m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7b69774ff9924472a46205284a314f95 : SUCCESS in 1h 53m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a54c3f32ebc43a5878800b536be244f : FAILURE in 1h 36m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/50631e3c7d5a4b71bc0bfd5a6c97384c : SUCCESS in 1h 03m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9b4aedcb70b48dcbd05492f4f122aac : SUCCESS in 1h 23m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/496342aaee694335bbbbccc42b6e2d99 : SUCCESS in 43m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/be8312892c9b41a9b14a5a955831e1e5 : SUCCESS in 1h 26m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/121b98cb8e0d4c2f9a3c9fa75cf45378 : SUCCESS in 1h 35m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b369c4e3b54a47b093fc403130cb1b0f : SUCCESS in 50m 57s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e37adfce4ecb4a3e9c86fcc61f488122 : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"313943c7be182fdab048c0db5459485eab41e270","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-01 22:01:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe6d3b6f0ed6a077435856fd88fe27954095da52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 22:10:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac006c9887ed483fa335d6ee9cc42212\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53ab8c3d45194214b1cd19f497b3f0b4 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5e9fc54aee549cd9b6b49d521dfa39f : SUCCESS in 8m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e42caa22f5c0d7c4c6c5d2a520e910012d383a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 23:50:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdb0e68b3d2a48afae93dc1f3c405dc5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f9990d35fc4b91bd7943e056d81e95 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58eb5d6397c44926bc885b66c73c8115 : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd25f1d8a47489688f95cb79bd9a6ed : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/13bd9e140729488088f4d70e58cd88cc : SUCCESS in 59m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bc0fb73c2c5940889eae0c9315a56a1a : FAILURE in 42m 05s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/41a871269f804d5e9a4563ba2bfd59a6 : SUCCESS in 1h 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50542c43dbca48688e45e9f421400f55 : SUCCESS in 36m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36a3912a3f9c44d5887d9fbd0e8dc1da : SUCCESS in 8m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/601f6646e097478396b9f129391c3ece : SUCCESS in 12m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ecd46ee4bd9543cca8382cf2f585ed10 : SUCCESS in 4m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/28730c48fbf1485fb649df5ac0dea103 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/71d70d2510b6458192bf10e1156cb196 : SUCCESS in 5m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/51e261a55ba24045a2229a11291276c4 : SUCCESS in 5m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0b5f115cead24cdbb5b47a1684464b39 : SUCCESS in 42m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/32943cc7af904cb3a9c44a9dd20565d2 : FAILURE in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/40c991d0b66c4fa5bc1ec9a79c8b546b : SUCCESS in 1h 28m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7a76c273c0fd49c6a0cb6b6e50883c38 : SUCCESS in 1h 43m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0ae053681c5242ea8fd9b0255d14a794 : SUCCESS in 1h 41m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/12dba68a16804ad58777f757824d0ad6 : SUCCESS in 1h 14m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5b9f883640224f16a971f6bece112cc1 : SUCCESS in 1h 12m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/35b52b0883264e52990a3fddf164b8ad : SUCCESS in 28m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/483ea7df3dff446fbc1da1791b2c0679 : SUCCESS in 1h 20m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8edd4e39ca4d49b7863ed034fe94170e : SUCCESS in 1h 13m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa0af60305494dbca298b9e472d91d6c : SUCCESS in 33m 00s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00601a1e238a45c1aeb0c54f46e878a7 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fde209d812e283ab3b1b79b6c5d979718d202984","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-02 03:48:49.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2c1970f034c541169c9ed41ae30e2632\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3b0f8bd192274ffa8dc1568b093e0e98 : POST_FAILURE in 24m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9536cb077d2623997a8125019a04a41b3e430afb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-02 08:33:35.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-910815-1/FC/ : FAILURE in 4h 40m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-910815-1/iSCSI/ : FAILURE in 15h 35m 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"d164d7bfc6f1cb4e97ea22533fe7b9d35fb6b0f5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-05 09:48:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/15/910815/1/ : FAILURE in 1h 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"79ebf3c3b99e305e92c427bcee7bf237e41b0fb6","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-03-06 14:31:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b02e085f1604ffabfaff79dc421a276f5c58bdfa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-06 19:22:09.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/13/.","accounts_in_message":[],"_revision_number":1},{"id":"8140ddb2a8c555e7c018a084a5d184729a67f443","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-06-02 20:28:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9ac472f23d8cac5c09c3a4e62b9a92ededad4c89","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-06-07 15:41:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7b0e44a19893c0c2c89abb68cb7e1b2b15ca5d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 15:42:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d7e84badaf94742253a8bc5b20f3b4941f0bede","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 17:32:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8dfe8bb449140748cebea0bc1bb0e4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3906f7af7a02415dacfc8a949f94ea8e : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eea2e8e2e5048a19ef6b526948ef8e5 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89eb9a96d1044be29f6eb6be4f2ef6f4 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/5f098c7f0e8a4856802427e9a3b0df9f : SUCCESS in 59m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1bb9bc9f12224290a2efeaa813e8652e : SUCCESS in 1h 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4545330b122a41dbbe58c16e829beaa4 : SUCCESS in 57m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/275f09efdd3444ac954b6b82296b3821 : SUCCESS in 17m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e96f091d93e416b805194a1e4411786 : SUCCESS in 1h 08m 08s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/69d18ab685234ca9b0d03f3e8e663cb7 : SUCCESS in 1h 44m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae44757d2f64416a8885e2883a4f7979 : SUCCESS in 52m 35s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9715a932362046b99a0ae4c0423c123d : SUCCESS in 1h 19m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0f2c1b7686a4c7417e1f9837885530750e0cc090","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 17:32:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9931646b747b263a57e4f589639255ba64c1099b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 17:34:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce4afd37e16c4c9488f83533226eaed6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b1c38d47884ce58957683c0dd986bb : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3702ea6c7f644aeb918c4639ac605915 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9330c6b6f17fb41d38fef38cbec49ad41b32714","revisions":{"c9330c6b6f17fb41d38fef38cbec49ad41b32714":{"kind":"REWORK","_number":1,"created":"2024-03-01 16:57:36.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/15/910815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/910815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/910815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/910815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/910815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/910815/1"}}},"commit":{"parents":[{"commit":"8d7e292bcd9ae610ac6341de2d00b8990252c24c","subject":"Skip sparse copy during volume reimage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d7e292bcd9ae610ac6341de2d00b8990252c24c"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 12:50:17.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-03-01 16:54:46.000000000","tz":-300},"subject":"Ceph: Fix restore backups to diff backend","message":"Ceph: Fix restore backups to diff backend\n\nNOTE: this patch is a squash of the major fix plus a followup that\naddressed some cases not handled in the original patch.\n\n\u003d\u003e Patch one:\nCeph: Fix restore backups to diff backend\nchange: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\n(cherry picked from commit 7b635086cbf4c374d73e6915636cb1e44b10b77d)\n\nThe current implementation of the `_full_restore` method in the Ceph\nbackup driver is causing restoration to fail when restoring to another\nvolume type. This is because the method generates an incorrect image\nname in the Ceph pool, resulting in the error message:\n\n```\nrbd.ImageNotFound: [errno 2] error opening image\nb\u0027volume-a91c11af-1147-4ac7-a5ce-61676736e076.backup.base\u0027 at snapshot\nNone.\n```\nOriginally, the format `volume-\u003cvol-uuid\u003e.backup-\u003cbackup-uuid\u003e` was used\nfor every backup created on the Ceph pool. However, after the change\nintroduced in commit Ia08c252d747148e624f8d9e8b0e43f94773421e0, the\nbase name format was altered to volume-\u003cuuid\u003e.backup.base when\ncreating backups from Cinder snapshots.\n\nThis patch modifies the current if statement in _full_restore() to check\nfor the presence of a snapshot_id attribute in the backup. If the\nattribute exists, the old base name format is used; otherwise, the new\nformat is applied. The if block statement in _full_restore() should now\nmatch the format generated by the _full_backup() method.\n\nAdditionally, src_snap is activated when the destination volume is RBD\nor if the backup is incremental. Currently, the Ceph backup driver only\nsupports incremental RBD backups, so incremental backups from non-RBD\nvolumes are not permitted.\n\nNOTE: This patch primarily mirrors the modifications in\nIa08c252d747148e624f8d9e8b0e43f94773421e0, addressing it as a partial\nbug fix. There\u0027s a consequential patch that further strengthens the\nbackup restore operations, which should be considered to fully resolve\nthe bug. These two should be squashed and backported together for\noptimal implementation.\n\n\u003d\u003e Patch two:\nCeph: Fix restoring old backups to a different backend\nchange: I074aeb222ca4566b630b5b76529454da4c8b493b\n(cherry picked from commit becf45c6cb5660fae59efe9c92ec7ebff0fcbce2)\n(cherry picked from commit 5d24ae2718520b5623eab23962097ecab5145614)\n\nThere might be scenarios where a very old backup, created from a Cinder\nsnapshot, was not handled correctly by my previous patch.\n\nThis new patch addresses this corner case by adding a fallback\nmechanism.\n\nIn Cinder, the base name of the backup volume is generated based either\non the volume id alone or a combination of the volume id and backup id.\nTwo different scenarios are considered here, each corresponding to a\ndifferent base name format:\n\n(1)\nIf the backup was made from a volume snapshot, the older base name\nformat is used: volume-\u003cvol-uuid\u003e.backup.base.\n\n(2)\nIf the backup wasn\u0027t made from a snapshot, a newer base name format is\nused: volume-\u003cvol-uuid\u003e.backup-\u003cbackup-uuid\u003e.\n\nIn both cases, the function attempts to restore the volume from a backup\nimage by looking for it under the base name. If the image is not found\n(i.e., rbd.ImageNotFound), an error message is logged, and then the\nfunction attempts to retrieve the backup volume under the alternative\nbase name. This is done because it\u0027s possible that the backup volume was\ncreated using a different format than the one initially tried.\n\nIn other words, the check for another base name serves as a fallback\nmechanism, which is used when the backup image is not found under the\nexpected name. The primary reason behind having two different base name\nformats is likely due to a change in the naming convention at some point\nin the history of the Cinder project. This approach ensures backward\ncompatibility and makes it possible to restore older backups that were\ncreated before the change.\n\nChange-Id: Ia3bca0e3d6dfcb29087cd680207319e5a663eb75\nCloses-Bug: #1895035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9330c6b6f17fb41d38fef38cbec49ad41b32714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9330c6b6f17fb41d38fef38cbec49ad41b32714"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
