)]}'
{"id":"openstack%2Fcinder~911329","triplet_id":"openstack%2Fcinder~master~I505b7cc12888d2373a0550b40bb945d75bb11067","project":"openstack/cinder","branch":"master","topic":"reno-eom-wallaby","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-03-13 08:38:57.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-14 16:45:08.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-03-13 18:33:17.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I505b7cc12888d2373a0550b40bb945d75bb11067","subject":"reno: Update master for Unmaintained branches","status":"MERGED","created":"2024-03-05 19:24:43.000000000","updated":"2024-03-26 09:53:59.000000000","submitted":"2024-03-14 16:45:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911329-reno-eom-wallaby","meta_rev_id":"2407a6a53d0f11acc8e4ce1795b06ec5cb992819","_number":911329,"virtual_id_number":911329,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 16:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-03-13 17:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-13 18:33:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-13 18:33:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 19:24:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-05 19:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-05 20:56:03.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-05 21:04:22.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-05 21:12:07.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-05 21:13:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-05 21:59:18.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-05 22:35:36.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-06 02:04:37.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-06 03:08:13.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-03-06 17:06:55.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-06 19:05:52.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-06 20:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-06 20:13:16.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-06 20:44:05.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-12 18:24:14.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-03-13 15:52:23.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 17:35:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-03-13 20:59:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-03-13 21:12:05.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 05:53:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-03-18 20:45:05.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"74f26eb27c3ee0227b6b2655cd2a65155279ffea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-05 19:24:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d52697ef9bdeb59ebe925e08cb6ffa43e5a31d0d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-05 19:24:52.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16592/","accounts_in_message":[],"_revision_number":1},{"id":"424696ca240420f3ca2d992c0414acabe4777ba2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-05 19:24:55.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16592/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7ccd1869e4812923fcfb4872413de0c7dfc1a482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 19:44:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc9102c3b7624a6b891e676ad598bb57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85daf46c49aa4f189c00c300cf8ee11b : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c38d9d31cff4dab82d86e600306e724 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cde0ab39f7ca452ea084d19bdbf9ee0b : SUCCESS in 10m 56s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/160865c3a5574e14a640bcccb5422184 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"086611e40fa386af88fb5b5bc8a89cfcb64632c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-05 20:56:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/911329 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e17d4f1cbffa95832a5e380a5a1ddecb6a7dda7a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-05 21:04:22.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9f514e6652364b2aba0b15787a38499e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/363f21f7a2364568a5ca5e32d8243bb3 : NODE_FAILURE Node request 200-0000002686 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"25d7d2e774c96f8ab0423ee886a75f1935919cd9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-05 21:12:07.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/29/911329/1/check/DellEMC_Unity_FC/768afcb/DellEMC_Unity_FC/1461 : SUCCESS in 1h 25m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/1/check/DellEMC_Unity_iSCSI/53fd96d/DellEMC_Unity_iSCSI/1492 : SUCCESS in 1h 46m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f5a9958e5deb97d0a490c2dad8b1dc4b25c6166","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-05 21:13:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/911329 : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0ecc8b3c0df7dcb3b1a96ef471b322ce197f88b8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-05 21:59:18.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-911329-1/FC/ : SUCCESS in 2h 34m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-911329-1/iSCSI/ : SUCCESS in 2h 1m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"111a0fe632cc2d3022789409c904915f0b67e9c6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-05 22:35:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/1/upstream-check/cinder-tempest-plugin-ontap-fcp/6d3a20a/ : RETRY_LIMIT in 13m 23s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/7ae1a07/ : SUCCESS in 1h 40m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/1/upstream-check/cinder-tempest-plugin-ontap-nfs/a384c3b/ : SUCCESS in 45m 17s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/230b6f5/ : SUCCESS in 44m 43s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e8724efab81471cbd7ce41c47ea3058c0dedaa79","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-05 23:13:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/911329 : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c6a436aacbdee498dcb1ea7e75cfc246bec4bb54","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-06 02:04:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/29/911329/1/ : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"75e8b72a31136de1cf66cd850f76be239469fd09","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-06 03:08:13.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/1/check/DellEMC_PowerStore_iSCSI/627afa3/DellEMC_PowerStore_iSCSI/1137 : SUCCESS in 1h 14m 19s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/29/911329/1/check/DellEMC_PowerStore_FC/21df0c1/DellEMC_PowerStore_FC/1306 : SUCCESS in 1h 37m 22s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/29/911329/1/check/DellEMC_PowerStore_NFS/6731cd6/DellEMC_PowerStore_NFS/821 : FAILURE in 9m 15s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/29/911329/1/check/DellEMC_PowerStore_NVMeTCP/6b5094f/DellEMC_PowerStore_NVMeTCP/689 : SUCCESS in 1h 55m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"679dc8ce6c0e269a04d992708e95f8bb55cd84fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-06 16:46:48.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"726d6fe9ebbbd4310a665267978d7c9bfba3fbc8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-06 16:46:59.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16600/","accounts_in_message":[],"_revision_number":2},{"id":"16deb417930123bef47442836340d894126b6276","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-06 16:47:02.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16600/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2e05fc51965273c977fc78786430a157ce960618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 17:03:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61d3ff15f22c4843ab041e1aba22f173\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ab923eb5f84175b896f9bbbd3e4fbd : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1896de358d264fc5a9bb0e5288f1dd62 : SUCCESS in 14m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6620c1f57da94bf6b0629509b31b0b6d : SUCCESS in 10m 33s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/95cbc25ee3e746a08175fd57fc4136ed : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"74099175370f701e593add8fbe41ea23a6e888e7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-06 17:06:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/29/911329/1/20240306213819/NEC-ISCSI : SUCCESS in 4h 01m 52s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/29/911329/1/20240306213819/NEC-FC : SUCCESS in 4h 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3496c79afe613badca64ba84bc7da9ecfa82ec4e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-06 17:48:51.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ca0274ac6b4346b1becfa3e2c882b953\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/89fd97129ac34f68b3005dd0de97b712 : NODE_FAILURE Node request 200-0000002703 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"ddd64f665164e88bb993b5a2f1d69da746143cf1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-06 18:15:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/911329 : FAILURE in 28m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4e713972e2793d73835ff0a7f7574295d639a7c9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-06 19:04:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/2/upstream-check/cinder-tempest-plugin-ontap-fcp/f6a34b7/ : POST_FAILURE in 1h 27m 40s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/f082b62/ : SUCCESS in 1h 59m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/2/upstream-check/cinder-tempest-plugin-ontap-nfs/01ca304/ : SUCCESS in 42m 55s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ae16d91/ : SUCCESS in 45m 46s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"128943b5ac512271208d4d48acfa724ad359aa42","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-06 19:05:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/29/911329/1/20240306212527/NEC-V-ISCSI : SUCCESS in 3h 34m 07s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/29/911329/1/20240306212527/NEC-V-FC : SUCCESS in 6h 40m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cdb0cf61c7223d2e57ca693c9ffa708651ad7be6","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-06 20:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/29/911329/2/fc : SUCCESS in 0h 42m 16s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/29/911329/2/iscsi : SUCCESS in 1h 16m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b5449c4f89f541f22cdc072f446fdd92be84d4aa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-06 20:13:16.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/29/911329/2/check/DellEMC_PowerFlex_3.6/5405de1/DellEMC_PowerFlex_3.6/1616 : SUCCESS in 1h 30m 10s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/29/911329/2/check/DellEMC_PowerFlex_4.5/51d7f24/DellEMC_PowerFlex_4.5/556 : SUCCESS in 2h 34m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"e27aaf36e2e63c54cb3bdc754c0141a8024ce5c5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-06 20:37:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/29/911329/2/check/DellEMC_Unity_FC/d8ad288/DellEMC_Unity_FC/1472 : SUCCESS in 1h 26m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/2/check/DellEMC_Unity_iSCSI/45acb87/DellEMC_Unity_iSCSI/1503 : SUCCESS in 1h 46m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b2ac44941dc9563d61177df008d41f8017a5998a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-06 20:44:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/14/.","accounts_in_message":[],"_revision_number":2},{"id":"58c7145bc7e7d71b16a924c74de92b35e6f92abc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-06 20:50:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/2/check/DellEMC_PowerStore_iSCSI/d3da683/DellEMC_PowerStore_iSCSI/1145 : SUCCESS in 1h 12m 23s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/29/911329/2/check/DellEMC_PowerStore_FC/6fd3ca9/DellEMC_PowerStore_FC/1314 : SUCCESS in 1h 38m 15s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/29/911329/2/check/DellEMC_PowerStore_NFS/46d49c9/DellEMC_PowerStore_NFS/829 : FAILURE in 3m 48s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/29/911329/2/check/DellEMC_PowerStore_NVMeTCP/3ebc5ac/DellEMC_PowerStore_NVMeTCP/696 : SUCCESS in 1h 53m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"f4172cffd7025e95252556b49214a7decc4520fa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-06 21:56:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/911329 : FAILURE in 32m 05s","accounts_in_message":[],"_revision_number":2},{"id":"fad4b55cfded7e154e6e0048d1e5a36a3ec844da","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-06 23:15:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/911329 : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":2},{"id":"de3149b645f66f4ffce6f5036e4d5fd4b55d3380","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-06 23:34:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/29/911329/2/ : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b30005cdfc6848c55bf036f1b6ca394d48110d8c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-07 17:39:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/29/911329/2/20240307232055/NEC-V-ISCSI : SUCCESS in 3h 14m 13s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/29/911329/2/20240307232055/NEC-V-FC : SUCCESS in 3h 18m 01s","accounts_in_message":[],"_revision_number":2},{"id":"0bb2b8b6b4499d9207b1893736d33d8b13c4759e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-12 18:24:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"102f2933a85e551de40ad6f9776c763d6d75a107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 18:24:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a53259e16b30d4facb5890e1d76cc259c3d581b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 19:01:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/380a8001e391427e99090e63265d94a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b952b8b408b44729ec546e7483b107e : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b7c400655245ce9c9fa99eb0438a2d : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bedfdb87b30d4b018bca6897417bd3a9 : FAILURE in 14m 03s","accounts_in_message":[],"_revision_number":2},{"id":"ab5213b41c68934e45554ca137e4607a9ef1ad56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-13 08:38:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"23a32b3a30ad95a11e764fb7ed3f914d68bb2684","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-13 08:38:57.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16625/","accounts_in_message":[],"_revision_number":3},{"id":"cd6c675d13ede7933f164701c529ecd699593607","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-13 08:38:59.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16625/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"bebf168b9edd973d6fb2cb3cd82fa445613e6753","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-13 08:44:48.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1b47435ce94d478aa91af5dc1665f0aa\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ed90eb2b0e194396a46a98fc28abe21c : NODE_FAILURE Node request 200-0000002745 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"f2e903d2c3c54a4c340dec562aa8a0ac51550b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 08:52:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b31520e7c89a49d6ac3505301114009f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6448608e55a24bc8942afe8f9728603b : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/934ab9f6390e461e827ac7924f1c793a : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ae90ede03234aa6848b8d699f857899 : FAILURE in 11m 05s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/764656a9eb0c44d1a330f627b9b3a74b : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cd086bd5abd05603387c4fc2938bd0cc453bd3f2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-13 09:36:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/911329 : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":3},{"id":"582e94f260aec68461a3be372f83759c08bd7a84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-13 10:17:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d98673a4820dd05f65d15175ed6ec008e57a0927","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-13 10:17:34.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16627/","accounts_in_message":[],"_revision_number":4},{"id":"50216763b845ab5814b71ef6480dbd3704c30f0a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-13 10:17:35.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16627/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"eff8e39900c6a3cf941cd85944af26751d944c4d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-13 10:23:52.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/dfc3a8a8f4fa4f21bb6671ab2e4cdda8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8f3c43ab974140818d1959eadafa8d8a : NODE_FAILURE Node request 200-0000002749 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"4dcc931fd5b40263a14b9b6e873a9b5c25399ae4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 10:34:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b40034f79da49ae9e71edc21a9e5a50\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bffac0bd7d344924a30470f062e3544d : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e3189f5959d4cac824f3abc97a9e2d8 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/88d46727ded942f99ffea36bc1de81d6 : SUCCESS in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f650f86141932d07b489f2714e489bf74fd16c66","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-13 11:19:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/911329 : FAILURE in 19m 01s","accounts_in_message":[],"_revision_number":4},{"id":"c79490271b80f67147c4f0867bba1765bef9e414","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-13 11:51:32.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-911329-3/FC/ : SUCCESS in 2h 26m 53s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-911329-3/iSCSI/ : SUCCESS in 2h 20m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"e9c4154efb46a25f96299d813bbdf5a2a0c851a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-13 11:57:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"933af93995a6d4a98532051da6ea0cd533e57a9e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-13 11:57:44.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16629/","accounts_in_message":[],"_revision_number":5},{"id":"68736643016ab81982e2fd0e5cfa9d80e0cd0413","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-13 11:57:47.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16629/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"ffd38ce282c512c143efaf8334ef2de1152e57af","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-13 12:03:28.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ede2c43227d74105a68ab8bcf5748114\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/85c5bc4ac5e14a54a48f3430e802027b : NODE_FAILURE Node request 200-0000002752 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"64e4ff56165cd73972e8886520a597e2751479f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 12:16:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/089cd3e7a97841dc8bb5b32cf9850b37\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4a3316ed59f4db5bf3cbe971b892155 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2e5e3529c424df98f980c8c48c5ae40 : SUCCESS in 16m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/612770abeb6241adbbb04d33eec3d90a : SUCCESS in 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"03e3972b36972002491543dd24bee95b912e5863","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-13 12:28:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/911329 : FAILURE in 17m 24s","accounts_in_message":[],"_revision_number":5},{"id":"37b717dee2572768e0470aa89c43107bf0b9b177","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-13 13:24:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/29/911329/5/ : SUCCESS in 1h 19m 00s","accounts_in_message":[],"_revision_number":5},{"id":"12cabb2eafc5976bdc7e2c1dc4769a06d004d9ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 14:07:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/001de1b9717b40be844cefc852d0e59c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cabe3123eab455399ba0b637937427b : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d51031543394016844c14eb27bfe738 : SUCCESS in 13m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/add4145eb68c475d9297ce2a50b3a18d : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff88c7ce0a340fdacfde81532030783 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/59db46969f884a07a088a59aade443eb : SUCCESS in 56m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f0564f2eacaa4c22b17b23c6f136f020 : SUCCESS in 1h 05m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d763ddd13a24b209b6107c1c042f3a9 : SUCCESS in 1h 45m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a0d533a0e254ba5ad58195d13c3c677 : SUCCESS in 46m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46aee56383c24119aa73d34b5dcaa427 : SUCCESS in 12m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59df3bbee1634141be34220e4e9ee3c0 : SUCCESS in 18m 10s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bca13764b6ee49549b19c40c8d6434a1 : SUCCESS in 11m 48s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a77e6e97c2d443c8bb80d3c72a1ef892 : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09b813c21ebe4509a2a281320b922824 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/739ca96b0bc640dfa0f766e24785f3af : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/db3df431945b4ea38546c2cd0baecb78 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/44e38f9f9ef242f0955e09be5599c420 : SUCCESS in 42m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63c4757cfd9f42158dfb753b48af1e74 : SUCCESS in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a198ee7a6390415486f722a5d1398476 : SUCCESS in 1h 23m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/37e735d2cc194313addcaf8a2b9dfd80 : FAILURE in 1h 49m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1649647e444b446280279a6f7a7cb370 : SUCCESS in 1h 50m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b244d57506e440f4a5d679ffe1d61a4e : SUCCESS in 2h 03m 43s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d815abc497ea4de89c3c6ff0d451ae6f : SUCCESS in 1h 10m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b603e3545f1f447d8cbc2f0e9b8372dd : SUCCESS in 1h 29m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e189cfedfad4737a67925f1454664d9 : SUCCESS in 36m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0bd12559706b4572bca37732d136b863 : SUCCESS in 1h 25m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4435776d4ff2431f9a8b6dec119f5a30 : SUCCESS in 1h 35m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8757d2af633b482594c2cf24ebfc2a48 : SUCCESS in 1h 38m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27e4d5be5ab54f86bac2a04998921ff2 : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":5},{"id":"1fbb0274b21a1be7ea5a3cabb58b64a60fb0cc02","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-13 14:17:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/911329 : FAILURE in 1h 15m 15s","accounts_in_message":[],"_revision_number":5},{"id":"ab5ceb23897275b9c3eb8ae13779da3c37d97fa2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-13 14:49:55.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-911329-4/FC/ : SUCCESS in 2h 55m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-911329-4/iSCSI/ : SUCCESS in 2h 12m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"43902a9def49006dc2e0d72ba3f5a090b5b97c24","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-13 15:11:50.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_Unity_FC/c627401/DellEMC_Unity_FC/1490 : SUCCESS in 1h 19m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_Unity_iSCSI/6b86def/DellEMC_Unity_iSCSI/1524 : SUCCESS in 1h 36m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"938d64fbb038a24d93864837c82093268309b382","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-13 15:17:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-fcp/3d1ce93/ : SUCCESS in 1h 03m 41s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/d567b4b/ : SUCCESS in 42m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-nfs/f1f63a5/ : SUCCESS in 41m 04s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/b3b8f6b/ : SUCCESS in 41m 23s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"a936a29859778f7047f9b4338eb6def99c6bde89","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-13 15:36:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/911329 : FAILURE in 1h 15m 42s","accounts_in_message":[],"_revision_number":5},{"id":"fe2eb8579e88541c0c3f7aa75c10bcaea2f222eb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-13 15:52:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/29/911329/5/check/fujitsu-eternus-dx-iscsi/5aa252f : FAILURE in 3h 51m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/29/911329/5/check/fujitsu-eternus-dx-fc/1e687f0 : FAILURE in 3h 52m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c30176208bc5aeceb230fc88352840a8d1631503","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-13 15:57:57.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerFlex_3.6/34ddde8/DellEMC_PowerFlex_3.6/1633 : SUCCESS in 1h 45m 59s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerFlex_4.5/d4a4e5d/DellEMC_PowerFlex_4.5/572 : SUCCESS in 1h 49m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"7ebfc266fcee1dd6199bf13f8ec67d529187a115","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-13 16:00:06.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/29/911329/5/fc : SUCCESS in 0h 41m 50s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/29/911329/5/iscsi : SUCCESS in 0h 56m 36s","accounts_in_message":[],"_revision_number":5},{"id":"32551ebe9c8b51a678df5be96c43a7776c448336","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-13 16:20:51.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_iSCSI/0be0306/DellEMC_PowerStore_iSCSI/1151 : SUCCESS in 1h 09m 30s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_FC/3ba4af7/DellEMC_PowerStore_FC/1321 : SUCCESS in 1h 36m 45s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_NFS/b2274c0/DellEMC_PowerStore_NFS/839 : SUCCESS in 1h 03m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_NVMeTCP/ed0f909/DellEMC_PowerStore_NVMeTCP/701 : SUCCESS in 2h 13m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"27906c3578179fe8ae4717c0c705a4cdbddbf481","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-13 17:22:43.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-911329-5/FC/ : SUCCESS in 2h 23m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-911329-5/iSCSI/ : SUCCESS in 2h 28m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"f36c2c4d4496ed69c297b238eefa22268f2102da","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 17:35:16.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3249fc5ebfcb1b3bbb9bbbfc40ae7741ca79b8ff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-13 17:38:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/43/.","accounts_in_message":[],"_revision_number":5},{"id":"2a8ae37c5ce90eb24994ccae9a6324e0c4827712","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-13 18:33:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"23c99e87f9fecf840fb0edcd2c0dc04f887f7f16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 18:34:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"20d11e9ee8dee8d68e6b9743ab34a890331f4211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 20:14:04.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/152b7aefd7ac4699b6e4606e7c295909\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a949fda58f4cb7b11ba451cdc2669e : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8b407c145f54244acedc06cb7fa8ce6 : SUCCESS in 10m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6394b0de39464099bd5b6f88f45119ae : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf8f1defca6945cc9f6e0dab23c43643 : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b7504cd34b5c49e187600f03db7fcfb0 : SUCCESS in 1h 08m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b8c13c15a58c4407bae620d5728bc681 : SUCCESS in 1h 00m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a4e1c4398ab4a169dc3f9a1608d4d1e : FAILURE in 1h 28m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f534f4ce53884fe89f2b099ad706f1b1 : SUCCESS in 1h 01m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65c93968349d4ac78504f54f308961c3 : SUCCESS in 12m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f17acb3796ae41e290e1d815fe7e5907 : SUCCESS in 1h 13m 54s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bdb82d78d7084c0d8c3f40c624d30429 : FAILURE in 25m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbd974fc652e42f4a59a869b04b0c284 : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":5},{"id":"f0a40b1917540913a6cf175dc1db27a5bc60aa36","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-13 20:59:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a1df60151021a35502cd9841b08fccdb4949a205","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-13 21:00:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dde4dee85828bc17fa596dff49332adf5d62a14a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-13 21:06:21.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/65b62d22fbb248fb93bb5d7de15a18a4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/09374f0b15d24cda97111d062ca83cac : NODE_FAILURE Node request 200-0000002759 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"9ec48a7d14c5b4fb5baae5130c948ffec0ad300a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-13 21:12:05.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerMAX_FC/62405c0/DellEMC_PowerMAX_FC/337 : SUCCESS in 4h 37m 30s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerMAX_iSCSI/4b2ccb6/DellEMC_PowerMAX_iSCSI/357 : FAILURE in 6h 01m 00s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"2bbf7017ddeb9006b62d3e5cf9834bd4028ed7fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 21:25:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f9754c23df8436db6585b851c9c3f53\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6899f005dd544d3eb274632916257098 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11d838ce568b4806afb58a34e4d4b622 : SUCCESS in 24m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/101c5beb309247f6b7bd6e268a5d6d01 : SUCCESS in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c54d29feab638b07d40fae24e0f66823dabfec86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 23:16:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2991e98d5c874b94a29ec5763ca06590\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f45d3a1cd3a4d0a9ba6489d612b8325 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b3f8129b8d24441a81e7d1e73bb7b0b : SUCCESS in 11m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/468709f9aff240dbafe2f6fd0b37ffa0 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3330ae529cc647b8a43374c4fb4872ed : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/47c4002bdb77485aa1f71a5405c7c73c : SUCCESS in 57m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/48c246564def4cd590b5a4658a6bdc7e : SUCCESS in 55m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19eded948f90438fb4cb5d7af60eed2a : SUCCESS in 1h 28m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cec368c8de84cc69aae157ce531b58f : SUCCESS in 56m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38d138a3dc80454183b5d08fa16b72f8 : SUCCESS in 11m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4161e6138abf48e593c4ba5e67483ebb : SUCCESS in 17m 22s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bb2322e27a134504847e6e40bd4ac1af : SUCCESS in 13m 10s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f6fe129b96d14c0fb3cd1ffa0eb48169 : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b10a345c302f4d28ae2ec76e758d71cf : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bb47235f3664a2f9e2116439a53e8a4 : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/459395ffa6664f2a93ff9d2bd2be2a2f : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b442abac9e4947d68a1576b339a5a11e : SUCCESS in 39m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d6f3152ce87466a8df711f31a518063 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d83753faf6be4d47b3feed4ac663fbde : SUCCESS in 2h 09m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/66b55f61703a4376963d5bbd9d48d5f0 : FAILURE in 2h 11m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9667a650217b47229e056ff8ede662d4 : SUCCESS in 1h 31m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bd3b848e98cc43028fbbeaf8f5c1f4bf : SUCCESS in 1h 41m 46s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/054d2b73c54144fe9e423a1c1a2956a0 : SUCCESS in 57m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/899571399c7f4b61b8d0883dc8febccc : SUCCESS in 1h 35m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/03592ddbae374979bca10917235e46f6 : SUCCESS in 45m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c8e95ddc273847969d614874a89d5895 : SUCCESS in 1h 08m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ede2275769db4c9ebc328951d518074e : SUCCESS in 1h 22m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/edb4b6eadad84f2d90db2acf44baa480 : SUCCESS in 1h 36m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55163a96f04b498292ff44896bf0d699 : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":5},{"id":"2ad3a3f343523af77258a2dd42d2533b9ef430b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 23:17:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"253b30e7d8273124ad8895534c52a9164cb76f9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 01:06:51.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b80a6083ac84f409db3ad8d4963d3b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d708ebde88f4391b7f99d78b2050f5a : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/471edcc9edda474e92ee5cb3c4d4bb5c : SUCCESS in 8m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6043ad45d9da44d6b141549621b1f616 : FAILURE in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96435af5f4b643e887288c4c61486719 : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/e62abe9ee3944b4d92ac76ed84e6a27e : SUCCESS in 45m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b7faf5ec5e17457381e30e74818185bc : SUCCESS in 39m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1feb19195b3e4df4bb5d7c4b0b463225 : SUCCESS in 1h 17m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf405355cecd431aaea73cfa92e4bcce : SUCCESS in 1h 02m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ea11b9eacc6430b924dd0383e638a31 : SUCCESS in 11m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d961b759a1164a5998057ddbf3a13320 : SUCCESS in 1h 25m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff9cc6526ae34bb3b3df28bea0d4f62b : SUCCESS in 1h 49m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b55128c8275342a8aaa347ede2e46edc : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":5},{"id":"a9e5dcc4388393aa65c87131915e5118a50dad7d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-14 01:10:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_Unity_FC/57b3c9b/DellEMC_Unity_FC/1497 : SUCCESS in 1h 21m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_Unity_iSCSI/82b543e/DellEMC_Unity_iSCSI/1529 : SUCCESS in 1h 26m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"3280df6ecea0fa3905e5218cac2e5983d0421e5f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-14 01:30:44.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_iSCSI/279890b/DellEMC_PowerStore_iSCSI/1155 : SUCCESS in 1h 15m 31s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_FC/6f05f2a/DellEMC_PowerStore_FC/1325 : SUCCESS in 1h 31m 19s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_NFS/d7f5974/DellEMC_PowerStore_NFS/843 : SUCCESS in 59m 44s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_NVMeTCP/e473bb6/DellEMC_PowerStore_NVMeTCP/705 : SUCCESS in 1h 39m 28s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"c46e24c710b728b2b90550473a151b86dc2ad58a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-14 01:44:29.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerFlex_3.6/57e87d4/DellEMC_PowerFlex_3.6/1638 : SUCCESS in 1h 48m 06s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerFlex_4.5/48e0801/DellEMC_PowerFlex_4.5/577 : SUCCESS in 1h 53m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"373e407acdaedd80b7e0b35d188f97e6def67e3f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-14 01:47:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-fcp/e457641/ : RETRY_LIMIT in 9m 16s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/2e3fb5d/ : SUCCESS in 44m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-nfs/0335a7c/ : SUCCESS in 1h 59m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3211d76/ : SUCCESS in 1h 52m 44s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"b9f1740c8079013f14fb2e6af71386b4f251de2e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-14 03:03:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"97b56878a079db74420856d011585f457c8b8833","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 03:06:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c8bc64485d3d186135f4c54c97cd3aa763eb621b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-14 03:09:59.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/40837d7b6efc43b8baac787bd154ff48\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c57ae61bf78848709bff7d6d3f95bfde : NODE_FAILURE Node request 200-0000002762 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"12052873b93a2c928b4cd234b10f53d34d6c091f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 03:31:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93e18d6d54c844c6992ad4a98c5f240e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/099f6077772540d2bcf47cbdbc596adb : FAILURE in 25m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc4a17d6f3754447bcd90c7483ceb01f : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/23c742aab30748b183e928d358cb3c50 : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3891db033963f06996f8f02b0bca0179fbacc1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 05:30:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c471618fabc4010bcae2c856b30d29f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfefd5317b3b4c1ba32d61bae983fb6e : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2f3d4740df54174b0cd4bb45c34e938 : SUCCESS in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3359949b78944fd397c5e5c6e104db21 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/010b94c2630747e28bf1109823e29f82 : SUCCESS in 12m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/3fd871435d31463089238caed6ea3a57 : SUCCESS in 1h 02m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/472491c502784542ae35341a488fc7e5 : SUCCESS in 1h 02m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3d439ed167a46889ee9e43578a0dcc4 : SUCCESS in 1h 23m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/687418e2f34348979c360fc83273483d : FAILURE in 1h 14m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88ddeadd59ff4005b04620a1e85d0f79 : SUCCESS in 12m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e869289dc418478889aa2d10b46f168d : SUCCESS in 15m 58s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d31e218854274ae0af30c79aee8ffe10 : SUCCESS in 13m 16s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3cda2868368446fb987cdec835481d18 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41760432deea4012be371a0fe7b3f7da : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/236ad6413b174eacb6ea9ba043797419 : SUCCESS in 5m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/29bc90fd015842e2b66399fe983a57d8 : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dda2641f36e1498281cf3d45b718af01 : SUCCESS in 42m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4ea9134a642b4f5fbdab8eed2e437a55 : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16f78b9d1ca04ed1b421632fb0f2a54e : SUCCESS in 1h 28m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c35045bb40f943e08b034fbe111df601 : FAILURE in 2h 09m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/881413825cad48c4b51119e56d17596d : SUCCESS in 1h 29m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ed86f448a3434bc5be2eff8d4270ddbf : FAILURE in 1h 51m 10s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55db1d73a0d74e9790905d08639922a0 : SUCCESS in 1h 21m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d165132492b548e4824c232ab39388eb : SUCCESS in 1h 24m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8532ba25822f4b1cb1290c58394830eb : SUCCESS in 36m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b873549de3146399d0cc3ee12b152c3 : SUCCESS in 1h 12m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ae27d93fa63b4e51bd3e9eb0e7e76d23 : SUCCESS in 1h 21m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f64289eed6f453f875ab1fa541ae383 : SUCCESS in 1h 33m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d82a8f5e016e41868bee7a70ece5d2a9 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":5},{"id":"60e5b83fead51df3b88a138ee50467d4a5ae5945","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-14 05:53:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3f3f477db947f5ed41e52098746c0c41a2909b2a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-14 05:59:31.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/0fc4ddd952a94fd3b13dd82b5e7502f0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1e7997f1ae4b4f21986587f1320e4d55 : NODE_FAILURE Node request 200-0000002764 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"d9948e977084c15f9b75e4ec4fc19b87353601a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 06:02:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20802bb7d2234d4f9063c00d747ff54c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96c033533e6c4b53bc233a59c47d2434 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f34523f0bf2843b9a2852c49f5ce20b2 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/88fc543bcec840bc88d2f42d0fa9c6af : SUCCESS in 7m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"33292d1ae3d2df4f8880055d81a4ba1461ff35cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 08:31:21.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65df60dd2c57409f8fad6080e86c5a3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac25837ea6d9446b9a7f07d43d25d4c6 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbed6b75f59c49d4b4191867eb9d9042 : SUCCESS in 12m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb1fcf8a946f416797de7f3a5bc77507 : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46b14fba89847789ba460eea441719c : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/0509603ed53248f6b65fa681fc4c5152 : SUCCESS in 1h 07m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2d5de16afee843b3b29f2b7b1f3be85e : SUCCESS in 48m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f7a05836e9a04ed78b2254d51ad8d289 : SUCCESS in 1h 20m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a392df1730c74d41bff8b569f10a41be : SUCCESS in 58m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f55d16297fa64c5380fd88dae5480d15 : SUCCESS in 13m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fff197b05451430bb4c722532e4c2158 : SUCCESS in 15m 37s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8a6217b0d1094cb7becf16f6dbfc585f : SUCCESS in 10m 11s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5937021dc8fb48fe84dab4752d5cf5d0 : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e6eee6995cb943499769f3603e2a8b92 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/254b294d95044d5dba4eb83257977a27 : SUCCESS in 6m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03ca60ce59c142f99a58edfa21f3658a : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/53a8285aded348918c9b4f7d3c187b90 : SUCCESS in 48m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a098711b842448bbb02ff554fd427ca7 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ec3900b5266480a9a4fbf602837e937 : FAILURE in 1h 52m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7b2b308641e74d09aeae6201fdd4edec : FAILURE in 2h 07m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6c47f47735914bc39a982a22443e0b1d : SUCCESS in 1h 36m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d7a459318aa7489d9e53a22652bf940b : SUCCESS in 2h 06m 59s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/909dbff00b924163a467861c668d66c4 : SUCCESS in 1h 11m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/88e924f202454386b178282d20900980 : SUCCESS in 1h 29m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1f37977737e34da1b78a886d437f1363 : SUCCESS in 37m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/125b5534282549268c640bac28d26bb6 : SUCCESS in 1h 25m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c8342d0ddec94e6e9809d08dd81495ee : POST_FAILURE in 2h 33m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9d28a1abb844a57b51573223045bcc7 : SUCCESS in 1h 42m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be5cfba873764eadbd7a582e3ed4689a : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":5},{"id":"053627f9c299450c4e9e253291cd889f3b1758a1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-14 10:50:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f5c55f6ac81f3319a0fce935a59c09e1b607c76c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-14 10:57:41.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8503f5708a804b3a976b97a8e96fece6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7d1a2609639f48b89265392bac031b3d : NODE_FAILURE Node request 200-0000002766 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"902838fa829fda76bab47530a022ae86bc45480f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 11:15:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abee15c09a93461dba702048e3031050\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0963d5e242b9429d81c7a358cc21aca9 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4bbc5b9d203d449f8f9afe51ca1e55bb : SUCCESS in 20m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d65fbb5fb79b4f7c8d8ad728087316df : SUCCESS in 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4efd8b9480c59c993ff4f399cc8dfd6a682c1d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 13:19:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac938c92e5de411885da297829921868\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e06a78af7d44ee8fcbbafcf694c51c : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a6f0153e3794e418f98a738453e7d40 : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37cc14f2cf7344828c4ae48c9e572852 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a383d99ae2fa4a348e6c3bc789384b0a : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/2f532cd6520a45ec8e447d015e8fd0c0 : SUCCESS in 54m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/56f09e70848d4b6ea076af920ccbd97b : SUCCESS in 1h 03m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e70d876266af45bd98c6cb37d2516105 : SUCCESS in 1h 34m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1e463dc6112418b99030844381d951e : SUCCESS in 1h 09m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e292a5ac46744f26aff420a1609b555c : SUCCESS in 10m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28bf3d2fe9564c66903ca0c68e80232e : SUCCESS in 15m 04s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f953b49d12764459a89d399e91f988f7 : SUCCESS in 15m 17s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ecd00ad8ea3d473da29925bbd09329fe : SUCCESS in 4m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b83103d126d4624aa14210cd46005ce : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/54424defa244407aa232bb69b2903654 : SUCCESS in 6m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/060d3264fd404104816f3d1845108949 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6dabb21d9ac5454c8583daef86c4beaf : SUCCESS in 39m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01a80072f1674da3aca48af0efe028ee : SUCCESS in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/84f8d4cff8a240d6911e5ce46ebc1971 : SUCCESS in 1h 57m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c8a4315f791a4195a082cfe2314586a8 : FAILURE in 2h 16m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9c1c1201529c47b897595c9f776d7105 : SUCCESS in 1h 35m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/378b4353f15f47a6bc27cd19c5f04718 : SUCCESS in 2h 03m 30s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f221361d5896451fbc14bbb8e57d89cc : SUCCESS in 1h 00m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5117d7814f6b461c8f1cd35ea6f3c5a8 : SUCCESS in 1h 35m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2602be091fb5450484f62cb4fca7b938 : FAILURE in 44m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1f771866c77145fa9953f73f9023bd48 : SUCCESS in 1h 26m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6657e317da494ccaa6c8e2594fb1b6a2 : SUCCESS in 1h 32m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2414ffce972748a9a06e959f539dccfc : SUCCESS in 1h 51m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0300cd9c788c490c822fca09a32fe64b : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":5},{"id":"52068af7de36e6fff7f2d6724a00e38aab35f884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 13:20:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"98c4254c4193e8502e69e9965b6f1f45147b2bc1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-14 15:10:57.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_iSCSI/0099aa0/DellEMC_PowerStore_iSCSI/1156 : SUCCESS in 1h 14m 49s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_FC/ed34e94/DellEMC_PowerStore_FC/1326 : SUCCESS in 1h 28m 19s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_NFS/5552a67/DellEMC_PowerStore_NFS/844 : SUCCESS in 1h 00m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerStore_NVMeTCP/2a2663e/DellEMC_PowerStore_NVMeTCP/706 : SUCCESS in 1h 50m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"f68100e4d126f8a9ec6abe24a3cc69ab53dff725","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-14 15:14:37.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_Unity_FC/bd7d148/DellEMC_Unity_FC/1501 : SUCCESS in 1h 35m 48s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_Unity_iSCSI/967d656/DellEMC_Unity_iSCSI/1533 : SUCCESS in 1h 54m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"2b02071f5af5ca2c605def1c286dc144718e3911","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-14 15:23:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-fcp/6e81fc8/ : SUCCESS in 1h 03m 52s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/a69449d/ : SUCCESS in 2h 01m 15s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-nfs/92a23a4/ : SUCCESS in 2h 02m 58s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/911329/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3bc3434/ : SUCCESS in 45m 17s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"afd90d52fdcb9cf5bc7700c4f43410d2e607c90a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-14 15:35:58.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerFlex_3.6/cc88c60/DellEMC_PowerFlex_3.6/1642 : SUCCESS in 1h 59m 06s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerFlex_4.5/27399da/DellEMC_PowerFlex_4.5/581 : SUCCESS in 1h 56m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"02357f33a2c6abe40135027c802429eeabe46d0b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-14 16:30:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/29/911329/5/20240314214520/NEC-V-ISCSI : SUCCESS in 3h 20m 39s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/29/911329/5/20240314214520/NEC-V-FC : SUCCESS in 3h 45m 11s","accounts_in_message":[],"_revision_number":5},{"id":"13edca4b1853d6618c84193490734539e6aa207d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:45:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ed225616ef435eb0ca2f15d3a3bf22\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f977cd80bf964f7b8c5a546c51d8642a : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9afca9415dc45f885cb4df71d2c17b6 : SUCCESS in 11m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/946db276dc0f422f88c290a5f7a452e6 : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f6515b26f848c598c437ac3a5966af : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c52b2ad517564933b82dc9cef7470b9d : SUCCESS in 1h 05m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f11d78c076994429a34c34e616adfd31 : SUCCESS in 58m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3eaa714ca5164fffb1b29aa3558bb758 : SUCCESS in 1h 34m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f17b8a88fa4340e68fa56ce0eb1f2dfa : SUCCESS in 1h 00m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe05e856c7af4151b23d5d1d1579587f : SUCCESS in 12m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0170c50770fa48bca5ef60fe87d1bc2e : SUCCESS in 58m 08s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5bf96d0e4234c3789f575add8289650 : SUCCESS in 2h 00m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ec619537b734d1580407feedbebbb9d : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":5},{"id":"0127f937408463d85bdb64fb75f3474d5e2d054c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:45:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"99617e7fe34dc714b967dba942b545ebd3e7c3e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:48:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11fe268a323142d2a04dea02c98c4e16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4a8323c6b4421db308951fc6aabc1b : SUCCESS in 1m 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66ae3159ae114c519ad66cd772c284e4 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":5},{"id":"838db053cdeb798a3d87cfc942ae7c70b839f2d6","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-15 02:16:01.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerMAX_FC/eb4cdd7/DellEMC_PowerMAX_FC/340 : ABORTED in 10h 30m 01s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/29/911329/5/check/DellEMC_PowerMAX_iSCSI/e5cae42/DellEMC_PowerMAX_iSCSI/360 : FAILURE in 8h 40m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"bbd9a3392fe6cfabb657441999dd6bef505d69f8","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-18 20:45:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/911329-1/Lenovo-iSCSI/ : FAILURE in 36:53.98\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/911329-1/Lenovo-FC/ : SUCCESS in 2:08:01","accounts_in_message":[],"_revision_number":1},{"id":"0bf605ab736e78d7c7f8f930aa8ccacfbb9974e8","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-18 23:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/29/911329/1/fc : SUCCESS in 0h 36m 16s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/29/911329/1/iscsi : SUCCESS in 0h 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a05d87ab9a9d6c05bfbc16b25b1695faaa632af4","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-20 13:28:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/911329-2/Lenovo-iSCSI/ : FAILURE in 5:40:04\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/911329-2/Lenovo-FC/ : FAILURE in 3:04:07","accounts_in_message":[],"_revision_number":2},{"id":"85186d9c9840f82bcaaad48e4672143278989105","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-24 20:16:32.000000000","message":"Patch Set 5:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/911329-5/Lenovo-iSCSI/ : FAILURE in 5:37:32\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/911329-5/Lenovo-FC/ : FAILURE in 3:02:44","accounts_in_message":[],"_revision_number":5},{"id":"2d2eff7879b7e8ff5614e279b331f5838fc41292","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-26 04:05:28.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/911329-5/Lenovo-iSCSI/ : SUCCESS in 2:04:24\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/911329-5/Lenovo-FC/ : FAILURE in 3:03:22","accounts_in_message":[],"_revision_number":5},{"id":"2407a6a53d0f11acc8e4ce1795b06ec5cb992819","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-26 09:53:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/911329-5/Lenovo-iSCSI/ : SUCCESS in 2:07:32\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/911329-5/Lenovo-FC/ : FAILURE in 3:02:47","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c0ff458b051ba414ff8e08d92b6e4a613fef0735","revisions":{"5889e07605d10b0a8c7b00ca2dbdf6b3e014c847":{"kind":"REWORK","_number":1,"created":"2024-03-05 19:24:43.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/29/911329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/911329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/911329/1"}}},"commit":{"parents":[{"commit":"a43cdaa85ca3ed00bf850e29dedb451a30316833","subject":"Merge \"[Pure Storage] Enable sync repl volume creation during failover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a43cdaa85ca3ed00bf850e29dedb451a30316833"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-05 19:24:41.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-05 19:24:41.000000000","tz":0},"subject":"reno: Update master for unmaintained/victoria","message":"reno: Update master for unmaintained/victoria\n\nUpdate the victoria release notes configuration to build from\nunmaintained/victoria.\n\nChange-Id: I505b7cc12888d2373a0550b40bb945d75bb11067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5889e07605d10b0a8c7b00ca2dbdf6b3e014c847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5889e07605d10b0a8c7b00ca2dbdf6b3e014c847"}]},"branch":"refs/heads/master"},"dc1cccd6627c8758f9a7368c011dde19cb3b5796":{"kind":"REWORK","_number":2,"created":"2024-03-06 16:46:48.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/29/911329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/911329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/911329/2"}}},"commit":{"parents":[{"commit":"a43cdaa85ca3ed00bf850e29dedb451a30316833","subject":"Merge \"[Pure Storage] Enable sync repl volume creation during failover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a43cdaa85ca3ed00bf850e29dedb451a30316833"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-05 19:24:41.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-03-06 16:46:48.000000000","tz":0},"subject":"reno: Update master for victoria Unmaintained status","message":"reno: Update master for victoria Unmaintained status\n\nThe stable/victoria branch has been deleted, so reno can\u0027t find its\nrelease notes.  Use the victoria-eom tag to indicate the end of the\nCinder project\u0027s maintenance of the Victoria series.\n\nThis strategy is what we used for the yoga transition, and was\ndiscussed at a cinder weekly meeting:\nhttps://meetings.opendev.org/irclogs/%23openstack-meeting-alt/%23openstack-meeting-alt.2024-02-07.log.html#t2024-02-07T14:06:09\n\nChange-Id: I505b7cc12888d2373a0550b40bb945d75bb11067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc1cccd6627c8758f9a7368c011dde19cb3b5796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc1cccd6627c8758f9a7368c011dde19cb3b5796"}]},"branch":"refs/heads/master"},"58bd8dab28b3986c7647d1a508964f9f4612e59d":{"kind":"REWORK","_number":3,"created":"2024-03-13 08:38:45.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/29/911329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/911329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/911329/3"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-05 19:24:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-03-13 08:38:12.000000000","tz":60},"subject":"reno: Update master for Unmaintained branches","message":"reno: Update master for Unmaintained branches\n\nThe stable/victoria and stable/wallaby branches have been deleted,\nso reno can\u0027t find their release notes.  Use the victoria-eom and\nwallaby-eom tags to indicate the end of the\nCinder project\u0027s maintenance of these series.\n\nThis strategy is what we used for the yoga transition, and was\ndiscussed at a cinder weekly meeting:\nhttps://meetings.opendev.org/irclogs/%23openstack-meeting-alt/%23openstack-meeting-alt.2024-02-07.log.html#t2024-02-07T14:06:09\n\nChange-Id: I505b7cc12888d2373a0550b40bb945d75bb11067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58bd8dab28b3986c7647d1a508964f9f4612e59d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58bd8dab28b3986c7647d1a508964f9f4612e59d"}]},"branch":"refs/heads/master"},"33fee477b58cb1534d6d0a1b34b87fe09658395c":{"kind":"REWORK","_number":4,"created":"2024-03-13 10:17:21.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/29/911329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/911329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/911329/4"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-05 19:24:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-03-13 10:16:23.000000000","tz":60},"subject":"reno: Update master for Unmaintained branches","message":"reno: Update master for Unmaintained branches\n\nThe stable/victoria and stable/wallaby branches have been deleted,\nso reno can\u0027t find their release notes.  Use the victoria-eom and\nwallaby-eom tags to indicate the end of the\nCinder project\u0027s maintenance of these series.\n\nAlso fix our override for the reno closed_branch_tag_re to include\nthe -eom tags.\n\nThis strategy is what we used for the yoga transition, and was\ndiscussed at a cinder weekly meeting:\nhttps://meetings.opendev.org/irclogs/%23openstack-meeting-alt/%23openstack-meeting-alt.2024-02-07.log.html#t2024-02-07T14:06:09\n\nChange-Id: I505b7cc12888d2373a0550b40bb945d75bb11067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33fee477b58cb1534d6d0a1b34b87fe09658395c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33fee477b58cb1534d6d0a1b34b87fe09658395c"}]},"branch":"refs/heads/master"},"c0ff458b051ba414ff8e08d92b6e4a613fef0735":{"kind":"REWORK","_number":5,"created":"2024-03-13 11:57:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/29/911329/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/911329/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/911329/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/911329/5"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-05 19:24:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-03-13 11:56:42.000000000","tz":60},"subject":"reno: Update master for Unmaintained branches","message":"reno: Update master for Unmaintained branches\n\nThe stable/victoria and stable/wallaby branches have been deleted,\nso reno can\u0027t find their release notes.  Use the victoria-eom and\nwallaby-eom tags to indicate the end of the\nCinder project\u0027s maintenance of these series.\n\nAlso fix our override for the reno closed_branch_tag_re to include\nthe -eom tags. Add the changed reno.yaml file to all relevant\nirrelevant-files lists.\n\nThis strategy is what we used for the yoga transition, and was\ndiscussed at a cinder weekly meeting:\nhttps://meetings.opendev.org/irclogs/%23openstack-meeting-alt/%23openstack-meeting-alt.2024-02-07.log.html#t2024-02-07T14:06:09\n\nChange-Id: I505b7cc12888d2373a0550b40bb945d75bb11067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0ff458b051ba414ff8e08d92b6e4a613fef0735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0ff458b051ba414ff8e08d92b6e4a613fef0735"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
