)]}'
{"id":"openstack%2Fcinder~912051","triplet_id":"openstack%2Fcinder~master~Ia8efa574b36d04132a8e384f6f39fb2eee164289","project":"openstack/cinder","branch":"master","topic":"fujitsu-driver-update","attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2025-11-15 13:53:55.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2025-04-04 08:14:45.000000000","reason":"Someone else replied on the change"},"37834":{"account":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"last_update":"2025-04-04 08:14:45.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia8efa574b36d04132a8e384f6f39fb2eee164289","subject":"Fujitsu Driver: Add mapping file support","status":"NEW","created":"2024-03-13 02:44:13.000000000","updated":"2026-02-02 08:50:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":38,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"489225a2a4b0e4b75b47be5c3f21ac45c6f14bc8","_number":912051,"virtual_id_number":912051,"owner":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-04 05:06:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 02:44:24.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 02:56:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-13 03:06:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 04:17:55.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 05:20:12.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-13 05:30:43.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-13 06:03:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 06:13:41.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 06:15:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-13 07:38:12.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-13 23:43:10.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-15 20:11:09.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 06:55:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 07:39:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 09:10:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-19 09:21:33.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-03-19 11:15:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-19 15:36:50.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-03-20 04:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-04-02 00:00:59.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-05-28 15:04:31.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-05-28 15:37:20.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-06-14 14:24:45.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-05 07:31:50.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-02 03:15:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-03-21 17:59:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-02 06:06:51.000000000","updated_by":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"reviewer":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"state":"CC"},{"updated":"2025-04-10 02:35:07.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-11-15 13:52:40.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"CC"},{"updated":"2025-11-15 13:53:55.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-02-02 08:50:44.000000000","updated_by":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"reviewer":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"state":"CC"}],"messages":[{"id":"33ad44fc3fd782dd079c3674fdf8fb27b1e26531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-13 02:44:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc1527ff4fc20a9887a540baafc438ddae051c25","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-13 02:44:24.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16622/","accounts_in_message":[],"_revision_number":1},{"id":"a25d322782700a8a5148c3eaf197cf760f6303c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-13 02:44:26.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16622/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0824713a46b4b6a93a131305f434ef1c81b73481","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-13 02:56:30.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/99b0e6c0af144a4d8543cc1106c4de0d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/366b455d57294f41b4cca2f61c11554f : NODE_FAILURE Node request 200-0000002742 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ab82f770069cae034a44c31f0974eee241a33190","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-13 03:06:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 17m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c7c450945fb205a58b34f94bac5430847a1b34c6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-13 04:17:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/51/912051/1/ : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"89a2a939278681d2ebcb2be7cf9f1a3eb82a4a47","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-13 05:20:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/35/.","accounts_in_message":[],"_revision_number":1},{"id":"ddc44d2c6289bd5232be4025e438f190f2c23cab","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-13 05:30:43.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-912051-1/FC/ : SUCCESS in 2h 35m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-912051-1/iSCSI/ : FAILURE in 1h 19m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"cef479a6ed7716d68f4e05718f14fc652d70a79b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-13 06:03:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/1/check/DellEMC_PowerFlex_3.6/14c1486/DellEMC_PowerFlex_3.6/1627 : SUCCESS in 1h 40m 22s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/912051/1/check/DellEMC_PowerFlex_4.5/a018fcf/DellEMC_PowerFlex_4.5/566 : FAILURE in 29m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"0d4a35f510e24e3256d865ba764dd198eb390af3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-13 06:13:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/1/check/fujitsu-eternus-dx-iscsi/3507d31 : FAILURE in 3h 20m 03s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/1/check/fujitsu-eternus-dx-fc/bc711c0 : SUCCESS in 3h 23m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c8f259eb70b1f776de7f0b794387836d2bafb3e4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-13 06:15:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/1/check/DellEMC_Unity_FC/ab218ce/DellEMC_Unity_FC/1483 : SUCCESS in 1h 13m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/1/check/DellEMC_Unity_iSCSI/c82abd9/DellEMC_Unity_iSCSI/1517 : SUCCESS in 1h 38m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"73c636b1246269ddefba395f52cf75a4694ba6cf","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-13 06:30:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fad9a9174bbf96a418b4da30671cbe2598269e71","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-13 07:38:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 50m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8bebe000b036003b0e97b3a64c0cf3b0b1112522","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-13 08:19:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 36m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d72ec19f6b8719338fc46879551f1e32cd5265f1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-13 10:26:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/1/check/fujitsu-eternus-dx-iscsi/a1c3eb4 : FAILURE in 3h 50m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/1/check/fujitsu-eternus-dx-fc/860eff4 : FAILURE in 2h 40m 13s","accounts_in_message":[],"_revision_number":1},{"id":"06ee30938b95a015af1ac5f0670be248df02f5b5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-13 23:43:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/1/20240314020731/NEC-V-ISCSI : SUCCESS in 6h 18m 37s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/1/20240314020731/NEC-V-FC : SUCCESS in 6h 35m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5de14bdccf85781161d4b0df04cf551b42f14ca3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-15 20:11:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/1/20240315200200/NEC-ISCSI : SUCCESS in 8h 34m 40s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/1/20240315200200/NEC-FC : SUCCESS in 9h 09m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7d97718c9a6f15e5ab233c9a6e561ce8f27b3768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-19 06:26:06.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"800e80e5bdea0c775befaa2b14c8b17c28c89269","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-19 06:26:13.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16672/","accounts_in_message":[],"_revision_number":2},{"id":"7a59bc135f6c0430d1b857cb835bb81490215590","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-19 06:26:16.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16672/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a9915bdbaa972c22c3911059a441ab927171a4ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 06:55:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3759acf9f9c4958880cfa964afeaa5b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35a2c2fc9baf4d6e9635c5bbb444b62f : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04851942f6fc41ee9da32d308915f141 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0bc59bf92a5d488ba197af3415f4f3a4 : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c879aff6612c726733bb2af5dfda3185a4a4cc4a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-19 07:39:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6292ca58c16c443cb5dace5265079441 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/158e5eaa015c42ed8cb9882f1b94d954 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"7c6d117922a3485a189ff7ed4dab412d7e61305b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-19 07:43:10.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/17a28570d5cc4c7182aabb3869460c42\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2898e2de3eff4294b3def5e9d61dfec0 : NODE_FAILURE Node request 200-0000002827 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"1558ab5f67352de43e100946bca99e3f975cf2f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 09:10:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6f57fb41a304ca1a0e9bbe3883bd833\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e41675379204105a1d8a523de7b1cee : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09459d1f47bf42e88b4dcca5487177c3 : SUCCESS in 11m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b10faa87f2344356ad59172d0334370f : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04cc2553f174289bc13d087e3107d81 : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/7241581d0501412b96ca923f0dd4350c : SUCCESS in 1h 02m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/55d2be94fd3049b993be2538fb8a7c89 : SUCCESS in 1h 07m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/771cd81285e74d2889687f32e39579f6 : SUCCESS in 1h 39m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb736d3149e74920a3c520d2b5909331 : SUCCESS in 55m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88e58c54386b48ca88f5bf611a23cf2d : SUCCESS in 14m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1c8cff783cf24662a9917f8af4cba032 : SUCCESS in 17m 20s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c3a4235efc844416ad442c63502951cc : SUCCESS in 10m 57s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d837273b2989410cb656d1fccc84b90a : SUCCESS in 5m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b762f03ace646e1be2b6bcc2b27de01 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/52b7c9daf3e94ddd844e2a1a0a85e2f5 : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5dfac09a15ad41529efca5440cdb141c : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3c7f7376718b461da42c4e9a409faa51 : SUCCESS in 46m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b3df47e34d144223bf8381a2a39df5ce : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6b53d176fd04b9c9c326ccc3e3b31b1 : SUCCESS in 2h 04m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/38bfeec6c2a145a7ba32a985769b53cb : FAILURE in 2h 36m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/25f1628b23d34870ac03296a21c2e505 : SUCCESS in 1h 36m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/580696bb753240ae8387ab5af573a3ec : SUCCESS in 2h 19m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1d02de16212443e380a6809c4285be5b : SUCCESS in 37m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e809a5cda397404a8b752c5b929687c4 : SUCCESS in 1h 20m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8a1f6ee925184769b9081f57837846ef : SUCCESS in 1h 29m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/71f4240f37fc4613b8da93a9a5fe29df : SUCCESS in 41m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fa3713c244c94362b0369d4c06e9149b : SUCCESS in 1h 31m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5bc13b404516445facb379f9591115cc : POST_FAILURE in 2h 13m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c79755508ae14b9797d100ad4efdca37 : SUCCESS in 1h 23m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24ed11ee343e411caf72bb5f02006be8 : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4c8bee66304df8df87562eed3467b28bfe6f3cf8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-19 09:21:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_03_2024_09_15_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"2e4e836fe5afd4fe29ceaee083b8b6662bb40b7f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-19 11:15:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/2/upstream-check/cinder-tempest-plugin-ontap-fcp/611e695/ : SUCCESS in 1h 17m 18s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/fea462a/ : SUCCESS in 1h 57m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/2/upstream-check/cinder-tempest-plugin-ontap-nfs/12f255f/ : SUCCESS in 44m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/37eb2d5/ : SUCCESS in 46m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"23bf489aab17707648de9da992441f427eeb7dd2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-19 15:36:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/2/check/DellEMC_PowerStore_iSCSI/302232c/DellEMC_PowerStore_iSCSI/1177 : SUCCESS in 1h 11m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/912051/2/check/DellEMC_PowerStore_FC/996f4a7/DellEMC_PowerStore_FC/1347 : SUCCESS in 1h 29m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/912051/2/check/DellEMC_PowerStore_NFS/237e55d/DellEMC_PowerStore_NFS/865 : SUCCESS in 1h 07m 30s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/2/check/DellEMC_PowerStore_NVMeTCP/9bb7834/DellEMC_PowerStore_NVMeTCP/727 : SUCCESS in 1h 46m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"357df666c16fe7baed470d33e0d127181b0906c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-19 15:53:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 15m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ecc76b5e0d029988ae6f04f1beaccbc50e745223","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-19 19:53:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/92/.","accounts_in_message":[],"_revision_number":2},{"id":"f510f33dc724e4671722fccc3ef2fb53a07186cd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-20 01:50:56.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/2/check/DellEMC_Unity_FC/cf44100/DellEMC_Unity_FC/1544 : SUCCESS in 1h 24m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/2/check/DellEMC_Unity_iSCSI/b65f642/DellEMC_Unity_iSCSI/1576 : SUCCESS in 1h 33m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"86b212409525547b0433c1cb55f0cf3814398865","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-20 02:54:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 50m 43s","accounts_in_message":[],"_revision_number":2},{"id":"73919c48881d79dcce7d8b16c81b1e0a2aeefc1d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-20 03:47:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 48m 22s","accounts_in_message":[],"_revision_number":2},{"id":"5062d00b32c8a6288369ffb3e6f0da0acf962f6d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-20 04:00:06.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/2/fc : SUCCESS in 0h 31m 37s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/2/iscsi : SUCCESS in 0h 43m 44s","accounts_in_message":[],"_revision_number":2},{"id":"0fb0d85fb65dea94c14d4e9c5037fb41df540c3d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-23 03:37:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/2/20240323100141/NEC-V-ISCSI : SUCCESS in 2h 20m 21s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/2/20240323100141/NEC-V-FC : SUCCESS in 2h 35m 52s","accounts_in_message":[],"_revision_number":2},{"id":"52279ce5829d8fb6336f7cf0ecfcff0cb47458c6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-25 04:46:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/2/20240325050529/NEC-ISCSI : SUCCESS in 4h 33m 07s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/2/20240325050529/NEC-FC : SUCCESS in 8h 40m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6dace1aae44aeaec09b51e99b7930f48c020a80d","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-28 04:04:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f5633621c409aa9e11b42eff0be2b4379620a02","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-28 07:50:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/2/check/fujitsu-eternus-dx-iscsi/bfcb3f4 : FAILURE in 56m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/2/check/fujitsu-eternus-dx-fc/8b56488 : FAILURE in 2h 21m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8fa01eb6dcb339bc27d57081f372c105c90c6829","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-29 06:40:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bec101b5a3b4fceb407344d803f0b4fdca29e126","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-29 10:37:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/2/check/fujitsu-eternus-dx-iscsi/cf6449e : SUCCESS in 2h 55m 18s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/2/check/fujitsu-eternus-dx-fc/c2b789b : FAILURE in 3h 51m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5b70a7ee2fb20d36a2877a67d447656e485a6482","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-04-02 00:00:59.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"dd820874f3dc94745fa119a3d5dc9844ec428001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-04-08 09:34:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7c3e79ca584414abf391815df4f7628d28c88ef8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-08 09:54:34.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/62dae8c45fea464190e6768f85f864e6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f23c4948776f4a9ea3f62990bf92cb6d : NODE_FAILURE Node request 200-0000002950 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"b27f31da81fdf4695e6a73dcd100379999824787","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-08 12:16:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/186/.","accounts_in_message":[],"_revision_number":3},{"id":"75c6982c0f119977e4217f899c106f9d5107f957","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-08 12:41:51.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/3/check/DellEMC_Unity_FC/3dc27e1/DellEMC_Unity_FC/1645 : SUCCESS in 1h 21m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/3/check/DellEMC_Unity_iSCSI/a77b722/DellEMC_Unity_iSCSI/1676 : SUCCESS in 1h 28m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d22bc1c7530d097fc8d5060450b1c322cb424832","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-08 13:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/3/fc : SUCCESS in 0h 39m 19s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/3/iscsi : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":3},{"id":"7d3a6f6c9709106c8cc573e828197e9e1ff960e3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-08 13:34:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/3/check/fujitsu-eternus-dx-iscsi/986e286 : FAILURE in 2h 52m 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/3/check/fujitsu-eternus-dx-fc/275c60d : FAILURE in 3h 54m 13s","accounts_in_message":[],"_revision_number":3},{"id":"9a1b0fbe07545eac8a6e95229da9bfc00daca499","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-08 17:48:46.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-912051-3/FC/ : FAILURE in 7h 48m 0s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-912051-3/iSCSI/ : SUCCESS in 3h 22m 45s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"997d5771ef577740c97fd2cc343fc5fccdba2bd1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-09 07:38:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/3/20240409115009/NEC-ISCSI : SUCCESS in 4h 38m 17s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/3/20240409115009/NEC-FC : SUCCESS in 4h 47m 56s","accounts_in_message":[],"_revision_number":3},{"id":"29f7bc4a67a975c1591600ea0f305a2dc31b5431","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 06:33:09.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"9dc9a53d67928b7027a81f163b2e39b2dca07d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-21 03:54:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ec18b70fb8192c3ec12a856c85ab5d98418e5aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 04:06:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca9f4026c9c34bcb936130e0ab2b514f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/daf9ce89611d4a558413ee755d6303a0 : FAILURE in 10m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/173f923947e24a4a8cc60102c2d8f658 : SUCCESS in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"16cbe15c1a5481edaafbaaa370af81365d0d9c0c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-21 04:22:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 23m 07s","accounts_in_message":[],"_revision_number":4},{"id":"8392e82a02fcd97336112722f6c27b11259c0757","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 04:44:40.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d3b7fd9c968c4e9ea7d73ffc3e08391f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6c874aaf16fc4d9482d04f3da65b1df5 : POST_FAILURE in 25m 18s","accounts_in_message":[],"_revision_number":4},{"id":"3cfe9db7e5d207a064d362c1dad6deb20d3ec6b8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-21 05:32:11.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 34m 55s","accounts_in_message":[],"_revision_number":4},{"id":"a04ccffb845a7491b1d85c5da688607e571cf9cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 06:24:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7f1b6278c04c5496199fa8ad48a0b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1231eac262714ca58215cbd26d7a3fa6 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25b2bcaa97f44156a814df819e15cf11 : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df867ad62e1a475b8847b8898a7fc674 : SUCCESS in 12m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e75f70db35f0480b98e53b848656b428 : FAILURE in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58563ae7fe6046128ba0b051262c3da7 : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/dfe5164620ad4bd38b89e988eccddcaa : SUCCESS in 58m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8376e0b4b2d480c842df26a68263cf4 : SUCCESS in 1h 36m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3d7d57a1e7649b096285f63a4181414 : SUCCESS in 1h 00m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f6728b9857340b6955c56a3c236ec65 : SUCCESS in 13m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/30743f29c30446bd9773f6252c4005fe : SUCCESS in 17m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0e74b02c7bd141ceb5063a70d518cf2b : SUCCESS in 6m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/007120a9a0f7473286c93a050243e755 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8ba15acad23241f38fb23f1e38fc5ab5 : SUCCESS in 6m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9d42134715d74217941a8c183076c314 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f3815c55565479e82689007671e218b : FAILURE in 47m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/675ca7c1c69a4947b06facf836431c17 : FAILURE in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a42fc6de16a4c8eb7ef1f456a116b27 : SUCCESS in 1h 39m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/278a1d78984e4891bcdcf97fe3a368c7 : SUCCESS in 2h 12m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cebe8a29d78a4005b0c46cdda4f32da2 : SUCCESS in 1h 41m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/35e2a7f5b8514db087e8665f2bd47f6e : SUCCESS in 1h 46m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eaf9ecb9cbb74978ae1f183aa100ccf6 : SUCCESS in 40m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4325333744c44db993eef9d342466cf8 : SUCCESS in 55m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a85ef1b955e485189b5cb978f45face : SUCCESS in 1h 52m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c30e7faf5c2a40efbf4773a1d78c73c0 : SUCCESS in 51m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6a609123fbb14ff482ce95e664b25c13 : SUCCESS in 1h 13m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9f174f1f81f548d993ad27c74e0ba654 : SUCCESS in 1h 06m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6defd053542402c8e1d478e29e81c15 : SUCCESS in 1h 36m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0dbc84dbd364bb282062a0c7f3c147a : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":4},{"id":"aa490309a03a1cb7fe5da86623af5b569bd6205e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-21 06:36:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/4/upstream-check/cinder-tempest-plugin-ontap-fcp/734e150/ : RETRY_LIMIT in 3m 54s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/e96cee8/ : RETRY_LIMIT in 3m 58s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/4/upstream-check/cinder-tempest-plugin-ontap-nfs/d315fbd/ : RETRY_LIMIT in 4m 08s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/cab9815/ : RETRY_LIMIT in 3m 38s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"1e393fb2ab25623628c7d96dcea06c68337a014c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-21 21:41:21.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/4/check/DellEMC_PowerStore_iSCSI/7cc1f0a/DellEMC_PowerStore_iSCSI/1360 : SUCCESS in 1h 18m 47s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/912051/4/check/DellEMC_PowerStore_FC/f845b55/DellEMC_PowerStore_FC/1543 : SUCCESS in 1h 06m 23s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/912051/4/check/DellEMC_PowerStore_NFS/9f306b4/DellEMC_PowerStore_NFS/1054 : SUCCESS in 1h 08m 40s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/4/check/DellEMC_PowerStore_NVMeTCP/3e488a0/DellEMC_PowerStore_NVMeTCP/908 : SUCCESS in 1h 39m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"c0e732509f15ab101138505e9ed66a68e38ff45b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-22 00:04:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/448/.","accounts_in_message":[],"_revision_number":4},{"id":"f2050e8cea6bb84d80c012320ccbc461702235b3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-23 03:16:17.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/4/check/DellEMC_Unity_FC/3d7b280/DellEMC_Unity_FC/1929 : FAILURE in 2h 10m 46s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/4/check/DellEMC_Unity_iSCSI/1feda23/DellEMC_Unity_iSCSI/1957 : FAILURE in 2h 35m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6c35974958edfb42c138566a7885fcd0332bac6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-24 10:05:02.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d001086108ea2f73aabbcb5185833b4958de5a15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 10:17:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1695297f0db34d0ea06f46f446723307\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d4c5a2c042645c685152f8016220698 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/557dca7d5e7942a489c391e3488f9246 : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"91606158027ef6b4c3e450b4ccc2644cc10ecf8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:36:45.000000000","message":"Patch Set 5: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0720b4ca08e44da89a17e0032227f5a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7537a62dc7c4409fba82a767eef50950 : FAILURE in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f22d2cb73a684d8fb4a917d84eb1d2bb : SUCCESS in 12m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ae1178154a74c619cf65bfff71a95cc : SUCCESS in 11m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f355fd061504d98b3a6fb347cc6092a : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b1af2163f634d2b99a60943a2f92ff7 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8505a346595946f199e1a1ff9fd68c2c : SUCCESS in 1h 06m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a987af9a5aad4e528400c814fb5383b4 : FAILURE in 1h 17m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b53c7152a854fe79cf03abb38a82741 : SUCCESS in 57m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eb1273191b8498683e70c729766e5fb : SUCCESS in 13m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/49df2eaaf11c452180b60d49c8c7167a : SUCCESS in 16m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a1157dc50bb4b9cb210f9f2c4d5cf82 : SUCCESS in 6m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f0237236de544b5afaa7ff210a88ad3 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65091a840f1d4766ba54e6ec2fa13060 : SUCCESS in 6m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3d6aed77c18a4b9584a07ae42bc37b31 : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/28c4d1beae034a52bcddcf4624d468c9 : FAILURE in 52m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9f5a8bf776f4294a8255eeb6f0d446e : FAILURE in 7m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f0ea1b7d041146009444992ef5c94b92 : SUCCESS in 2h 10m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5b948d862b8e402983752f5b4bad41de : FAILURE in 2h 21m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8fa3ec5a42b24bd4a5607a90ce206a44 : SUCCESS in 2h 00m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/68fec03c6c494811a5dfe996582b3725 : SUCCESS in 1h 56m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/15964787ecc54bafa17d43a38836ccce : SUCCESS in 40m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ae83494b8094a14a2dfc7ecc7844dd2 : SUCCESS in 1h 04m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/50686576c013443a8c0fe1dfe88cd5ce : SUCCESS in 1h 28m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/52592a2ef29147cc855646136c44ce66 : SUCCESS in 50m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9f0fc2a25eec406e88ddb91d4e7481b5 : SUCCESS in 1h 38m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cf8fd4a9655047e788349ab1d4c2e4e4 : SUCCESS in 1h 20m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/288bfbe8c376444cbcfa88c86c77d21d : SUCCESS in 1h 38m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dafd690199474772ab15bd1f2cba4e96 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":5},{"id":"2c15852dfcc8345a6ab357bb54d5c02652c844fc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-24 12:42:57.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3a03663ab8e8495a98674921216fb70f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4eaf9824022040599c6998eb77ca4f64 : POST_FAILURE in 25m 36s","accounts_in_message":[],"_revision_number":5},{"id":"a97c3d9956567eea64e54ffd511c8ac559a2a441","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-24 13:52:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 22m 59s","accounts_in_message":[],"_revision_number":5},{"id":"de4b1ee02a4435545287108eb5f40807641b716e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-24 15:29:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 34m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b16e1df4fd15ce9a1274f076a1210c2c0f6cd315","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-25 00:24:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/493/.","accounts_in_message":[],"_revision_number":5},{"id":"b661c6ccace287e6fc406b7a1f146602cab0b8de","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-25 16:49:38.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/5/check/DellEMC_PowerFlex_3.6/f4e59ca/DellEMC_PowerFlex_3.6/1447 : SUCCESS in 1h 41m 24s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/912051/5/check/DellEMC_PowerFlex_4.5/fb34cd3/DellEMC_PowerFlex_4.5/361 : SUCCESS in 1h 52m 43s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/3a4dc2d/DellEMC_PowerFlex_4.x_NVMeTCP/301 : FAILURE in 2h 56m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"87ddc96ad74c52f7b6c3c9cbc0b92d1414db1310","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-25 22:37:12.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/5/check/DellEMC_Unity_FC/9e181e7/DellEMC_Unity_FC/1966 : FAILURE in 1h 20m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/5/check/DellEMC_Unity_iSCSI/90e06c4/DellEMC_Unity_iSCSI/1991 : FAILURE in 1h 46m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"3fbdb1c5956bd510490dfc7007e69c9e27d69313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-27 09:50:14.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8ef49faf0436ae27b457e3dd73d4649d43c35429","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 10:12:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/350a06e61c944454848beb0cc5cd81fc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7359d2efe0124a29ab8152884290a8b5 : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0d40c0cf1c140e3bf221d5e3322ff57 : FAILURE in 18m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"21c26f1ddf1e133865704b55602a7f385849e417","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-27 11:12:13.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/744520e8870c40ff9aef6e5877204ef8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ff8fc47d85994e6daac3b8161056ddfe : POST_FAILURE in 22m 17s","accounts_in_message":[],"_revision_number":6},{"id":"cfd97b794172223a871f6fc538ae6360d408b3d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 12:09:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37c97b50650f48c99538c173dbb6922b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36e3ce4020643c79125a5c8020ac4cd : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78752012f923423cbe884a74836764fe : SUCCESS in 11m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2384b489a7904e2fbda8ed079e8c909a : SUCCESS in 9m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddc0581ed3a5442e9bec18f5ceecf6cd : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b0b7b7b92524ea4ba4d628ff7b792f1 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/f386503fd15d431ebad555d89ac39e11 : SUCCESS in 1h 05m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8134862019745cdae10dcbef4b7146e : SUCCESS in 1h 34m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8990908059494b309d6427b9e56b14aa : SUCCESS in 57m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aa64ca7f2ca4c6188a985001c1f772a : SUCCESS in 17m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95023c60f8454e3cbc69fdd01698f381 : SUCCESS in 20m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c7de656dac874ea681745fe18815e991 : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88236ed80f1144edb39637048c557857 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e6b2895d6acb4bc68b6bf9dabac2dda8 : SUCCESS in 6m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f82ad82607284970b5e932d467332bed : SUCCESS in 7m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2ecd1e84e7254134bf48a26699bc5139 : FAILURE in 32m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/efbdb6e4d1a442bdafe1858e3d786171 : FAILURE in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/39eb0cbe9e1a4eba9a99a4b234e2d82c : SUCCESS in 2h 07m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1a67724a4b194a4f8e919ba3838d9484 : SUCCESS in 2h 09m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1a867bc9be954097a178164a551d6b67 : SUCCESS in 1h 52m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3afa8083d6324b828324545809d5fbc7 : SUCCESS in 1h 51m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/81e91995dc884d1e866ab8ab2ed81e87 : SUCCESS in 42m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e73c3c5fd0604e889ca32ad0e39ed2fe : SUCCESS in 1h 16m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/57e374568e9040c9ae42ec61c1b77ad9 : FAILURE in 1h 39m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0d291cba431042b98af48298e77df110 : SUCCESS in 43m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c7460564c64045caa23c9904c94f99e2 : SUCCESS in 1h 33m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/111d038793eb4797be355ed413148f89 : SUCCESS in 1h 31m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2dd4026c0adc4d70a3fb62304f047a4d : SUCCESS in 1h 44m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab5e8b19c2754bffb5e8794eb22fe373 : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":6},{"id":"793fc82afb3b3af361ab73dcde9723601ec684ee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-27 14:37:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/6/upstream-check/cinder-tempest-plugin-ontap-fcp/e7cb9b3/ : RETRY_LIMIT in 4m 31s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/144328b030dc40aba892b183cb3ad13e : RETRY_LIMIT in 2m 19s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/6/upstream-check/cinder-tempest-plugin-ontap-nfs/9f1aa9a/ : RETRY_LIMIT in 3m 59s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d41dc74/ : RETRY_LIMIT in 4m 35s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"20e1308dbed6aa3214a3070b767bdc626999242f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-27 15:17:19.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_PowerStore_iSCSI/6c80e3e/DellEMC_PowerStore_iSCSI/1386 : SUCCESS in 1h 20m 51s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_PowerStore_FC/e470322/DellEMC_PowerStore_FC/1569 : SUCCESS in 1h 08m 29s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_PowerStore_NFS/814ac45/DellEMC_PowerStore_NFS/1080 : SUCCESS in 1h 07m 16s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_PowerStore_NVMeTCP/dafe679/DellEMC_PowerStore_NVMeTCP/934 : SUCCESS in 2h 00m 28s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"bbcc61398cde7f31d678f0d9dc467a38dbb0bae3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-27 15:32:54.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 36m 04s","accounts_in_message":[],"_revision_number":6},{"id":"0cc2d7bd94f7bcc855e8b65dccb2e7d69406a340","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-27 20:43:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/514/.","accounts_in_message":[],"_revision_number":6},{"id":"7eb71deb5262e8187eb2e5e083fa7133de872b64","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-28 04:48:36.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_Unity_FC/063bb2e/DellEMC_Unity_FC/1987 : FAILURE in 2h 10m 35s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_Unity_iSCSI/7ea7545/DellEMC_Unity_iSCSI/2012 : FAILURE in 2h 20m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"ab88c3dd2661c27eb6d759e5bfd2e04e195ad120","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-05-28 09:10:57.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"fcfa16ee24b7e9dae16f590a1bff57e707ec5fca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-28 11:27:41.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_PowerFlex_3.6/b26b787/DellEMC_PowerFlex_3.6/1468 : SUCCESS in 2h 05m 51s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_PowerFlex_4.5/0c98dc4/DellEMC_PowerFlex_4.5/382 : SUCCESS in 2h 12m 37s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/9fb978e/DellEMC_PowerFlex_4.x_NVMeTCP/322 : FAILURE in 2h 56m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"515fa5ef1ab8d1ee2a8287b0e5f7febfaac76949","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-05-28 15:04:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/912051-6/HPMSA-iSCSI/ : SUCCESS in 3:17:53\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/912051-6/HPMSA-FC/ : SUCCESS in 1:58:14","accounts_in_message":[],"_revision_number":6},{"id":"7fd254bdec9ce42a390b0b532127af8cfbe4a313","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-05-28 15:37:20.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/912051-6/Seagate-iSCSI/ : SUCCESS in 2:23:58\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/912051-6/Seagate-FC/ : SUCCESS in 2:17:31","accounts_in_message":[],"_revision_number":6},{"id":"6d3c68f06a8130b0fc4979e9520dc5b82091a9bd","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-29 10:00:07.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/6/fc : SUCCESS in 0h 41m 35s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/6/iscsi : SUCCESS in 0h 50m 49s","accounts_in_message":[],"_revision_number":6},{"id":"318e8389e3561c6b4a230014e80c6a08f267b5a2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-06-06 22:47:58.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/6/20240606224705/NEC-ISCSI : FAILURE in 8h 44m 02s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/6/20240606224705/NEC-FC : FAILURE in 9h 00m 50s","accounts_in_message":[],"_revision_number":6},{"id":"f44af6ab00d526a6f58f5c871747c704434dfe0e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-06-07 07:52:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/6/20240607131126/NEC-V-FC : SUCCESS in 3h 29m 11s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/6/20240607131126/NEC-V-ISCSI : SUCCESS in 3h 41m 26s","accounts_in_message":[],"_revision_number":6},{"id":"75a5099f14bbe95ce3d5301ca381d2e10b1eda7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-14 06:58:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1a1ab40aaae7f4f7cc8e930b3b7279a86fe0bf04","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-14 07:00:59.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"8460449458401db78dde7e50e7e567aaf4febaa7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 07:22:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb61b71d622498d822572744f48ef1b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c80c0f1f369744f7b68fc740bc1eaf7a : SUCCESS in 21m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb2f2987be3149d884f9a83b757c8377 : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e7d9c6ec1883ecd9104167ad3ced4baabc89ae64","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-14 09:15:54.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4a221b7da4c542de8b5bf8decc00a37d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/42e48af42a8845f0aa08d2631015ce96 : POST_FAILURE in 27m 44s","accounts_in_message":[],"_revision_number":7},{"id":"b64aaac756b3cd444cbb9a5f222d7348b0c240af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 09:51:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1560382c883f456594d64301cb043afc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80cbbd8465d42c184afa94901f9a6a0 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1209cf760465484c9ba372fe205c267a : SUCCESS in 10m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55f5448524884c7c8286987107821bc3 : SUCCESS in 11m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd2de3adf1904cea8badaa9f57306c7a : SUCCESS in 16m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e567309a3c040d4b0bb32ed0d0c55ac : SUCCESS in 14m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b9e2716787e547b6844dec92fec1ce94 : SUCCESS in 1h 04m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fff98359ed62407f9bf18fa47e74bf6b : SUCCESS in 1h 42m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/529a73419bcc4c3d93cceb4280c7040b : SUCCESS in 59m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b50a4b38d5a44842a387973999ce579f : SUCCESS in 23m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/70c272b29a364adf8fccc5d04e29893f : SUCCESS in 16m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/949ca8c398ef4e5781375048b8fcc4bf : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7e48da33c67c4d18b08344d836818be4 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78b488e521a448e787de8dcffffb60c6 : SUCCESS in 6m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3113df4fa872477e8f0f48d2b354f083 : SUCCESS in 8m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/261d95a9d8674db188a4a2fcb0d016d6 : FAILURE in 37m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/acd604de7b69460c8b81899e6af32986 : FAILURE in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f72cf0295c67448791bbffc89bc22bbd : SUCCESS in 2h 19m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8034115a192b43088ccf9c853d6f7b6a : SUCCESS in 2h 40m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/647ba011b67f4c11b72d073f6e091444 : SUCCESS in 1h 44m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/47b1baf3fcdd42b6ac649dfcdc9be65c : SUCCESS in 1h 37m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3c9a5db195e449e595ac557454a4db13 : SUCCESS in 29m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0867a2e3c3e347569a60e797679a5433 : SUCCESS in 1h 14m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/72a6a979a68249f4bd5020be380fae7b : SUCCESS in 1h 37m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5fa61bc1064f4abb96b8f9dc9563c4a0 : SUCCESS in 53m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/836c82d3fc6a481c8a911a3d96b61dff : SUCCESS in 1h 41m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b11bd9e3aded49099a5430c614671641 : SUCCESS in 1h 39m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fff4be18ad7b4c2ea3db4800ac3aa0c1 : FAILURE in 1h 31m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52c7535acd574aa1999700120c13ee31 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":7},{"id":"25ce7b4ea3c5e45809af25d85035ab187021b110","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-14 09:53:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c60bfd5d60daf33fd3f401709578c6e6125b6fa7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 10:07:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f65f2fc168c430190448f13c1278a71\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9afb600300534753bb9b276197cbf335 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e77226bada6a47c48c757527db02bae7 : FAILURE in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"91ffed8aeb0351697d2fac5f5f3eb572fac8b117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:34:15.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3c7860a56ad495b9cf2228c9127f93a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9af6eab137a441b986621abe3035ddd : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d929870e2f3b4a9baa9a874cc57dec37 : SUCCESS in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f83833d846d14f918450d8963fa50d2f : SUCCESS in 10m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3bd0c0d86844db199e74f9e84d83a3b : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1912a6582db4b4ebf9a27dc1a765d60 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/8d7d72ddbdaf41dc92d9f37fd63ae698 : SUCCESS in 1h 04m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd61089d14cb49d18448d7f53e359003 : SUCCESS in 1h 25m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fcf3c69cf804029abe21804e0a54031 : SUCCESS in 1h 09m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/683fd51ae72149be91b878c3e4823d8f : SUCCESS in 18m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/604821f592944b2db4ecba76bf64ae74 : SUCCESS in 16m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aa095264bc49499184d875623b875c3e : SUCCESS in 6m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4dd96acefe464563bd1f871e20368702 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9be0704aae524464bc064eb84506fdcb : SUCCESS in 6m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a9c457b241374715ae735a1896bc2bdc : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2a2e4f3d367c4c9fa29364f14d394da8 : FAILURE in 46m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/633532563fc14ae09f4770d08cd5a576 : FAILURE in 6m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/49915eda569449929f2b4037f9bb01ca : SUCCESS in 2h 02m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3878d1a2a52045e0ac49457ea13a2994 : SUCCESS in 2h 32m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/76b627b73a3146aa9e7045866be0916a : FAILURE in 1h 50m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f99af350776a4a8fabd8aa50e835fc4f : SUCCESS in 2h 04m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0376a3d2ee304c25959db98d721d6395 : SUCCESS in 34m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a3ee95c680642aeba3ba0fd2b74bf54 : SUCCESS in 1h 06m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/80f6c49fceb547df924a21e0f1d1df78 : SUCCESS in 1h 29m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8529fd242d794627a3cb75aadb6cf41c : SUCCESS in 42m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a314c4d079e428a84fe125bb42475a9 : SUCCESS in 1h 39m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0e662dc9297f469dbb15257a3f5521f1 : SUCCESS in 1h 24m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89c3adc38dd9490781a2e9901c7148b9 : FAILURE in 50m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c4be6fd1bdf4d1e93fe531df6142c57 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":7},{"id":"c70e30d21e146a367979ea2ac7a0d72da14e0712","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-14 13:23:32.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/91e1ab3aa69c46b0b17beac43276aa58\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/cac8629d2cca4be186d80cd506515f45 : POST_FAILURE in 21m 51s","accounts_in_message":[],"_revision_number":7},{"id":"c379255f26f2a1f1cda2c844e4a6aaf71f776e4f","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-14 14:24:45.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912051 : FAILURE in 31m 21s","accounts_in_message":[],"_revision_number":7},{"id":"cfef65fd53a32e3c5ef409b55903119a82e2be94","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-14 15:34:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 36m 10s","accounts_in_message":[],"_revision_number":7},{"id":"558c995d53e50b7e83372365b34bbe77d6be8c94","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-15 00:31:00.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-912051-7/FC/ : FAILURE in 10h 13m 23s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-912051-7/iSCSI/ : SUCCESS in 5h 48m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"a3f9faa38b6e7167804f5f19276cd250ba4e8661","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 02:02:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/117/.","accounts_in_message":[],"_revision_number":7},{"id":"f4261eed75c1ec80ec432e003e4cec01d8355634","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-15 03:00:55.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-iscsi/7d5f4fa : SUCCESS in 2h 31m 29s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-fc/39afdab : FAILURE in 3h 53m 29s","accounts_in_message":[],"_revision_number":7},{"id":"5dcaa742f69bdd31d42ecac932b7d7b256e889df","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-15 05:35:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1e800837604e4e29ac060156b9cf5770 : SUCCESS in 1h 19m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1a3cc45dd81042388b6e3d0612a8a648 : FAILURE in 1h 14m 41s","accounts_in_message":[],"_revision_number":7},{"id":"6b782f02dfa558d327aced74c82404ca7b3e6f4a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 05:36:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/603/.","accounts_in_message":[],"_revision_number":7},{"id":"6a67b4caea127dba37718f95be8dc8614846aaeb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-15 16:01:12.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerFlex_3.6/bb02244/DellEMC_PowerFlex_3.6/1578 : SUCCESS in 1h 53m 06s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerFlex_4.5/ea72f5c/DellEMC_PowerFlex_4.5/473 : SUCCESS in 2h 00m 39s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/3108b84/DellEMC_PowerFlex_4.x_NVMeTCP/423 : FAILURE in 11m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"f78a4f5d90d50f8430a8ac7497faba99111b374a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-16 07:18:49.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_Unity_FC/efe4671/DellEMC_Unity_FC/2087 : SUCCESS in 1h 28m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_Unity_iSCSI/5d10022/DellEMC_Unity_iSCSI/2110 : ABORTED in 6h 00m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"86349d637183b8653cc10da7ac4ec92b547170ee","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-18 07:53:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dc59f582e3d1c69e49526d8ca17ed599a170705d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 08:11:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a9a25adbb8a4a6c910e38ad32e16a66\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b3e965c570a48d7a2a305e747994192 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/65d9fb11613e480c985e9df50140cf6f : SUCCESS in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5647984571ebc76b68a54493223cec59e7689816","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-18 09:23:53.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8fafe69a3f6543bc92af92753c45c76a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/43b31f7981484881b5d918da0354fa9b : POST_FAILURE in 27m 42s","accounts_in_message":[],"_revision_number":7},{"id":"c2dfa26eb8478749d4c97fb503b973f926b20885","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-18 10:41:31.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5f71dd84b141430d85f9d81751af9029 : SUCCESS in 1h 19m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5e0ce9ee8ef746f0ad300a7a891cf69f : RETRY_LIMIT in 7m 39s","accounts_in_message":[],"_revision_number":7},{"id":"46c2156f3783b112d2b3857d36d0ce2017283c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 11:05:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba0176be216d427aa2358afee752674c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89673b31e8fa47a398b133396c500bf0 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46736b3ca12c428b884d8e0dfc85c25a : SUCCESS in 10m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/693f22d26e524a8a8070ff06f804d84f : SUCCESS in 10m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e8bf6dd2ed8b4cfaac48733ee7120632 : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04cbb773e7f45b6ad83cdf1fdb4a046 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/16e29b4889e04676aa08744ae5fef0d8 : SUCCESS in 1h 07m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c12a44e1906740198a62095b4d01cad8 : SUCCESS in 1h 44m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/183f8d8f3247404299ecbaab4f4ea595 : SUCCESS in 1h 16m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/459740a8c63b443187075383f791d77d : SUCCESS in 17m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ed2967d384f14c7abcdc58ca63aaa96b : SUCCESS in 16m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d04c88cea39c464aa5e78aaa824c865a : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ccf22b1a2c1849a48e647b20234755ce : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d72e0a7352a049a89257de8952785585 : SUCCESS in 5m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/959afd77d69847adbe3617ea7dfdf2bf : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/986917852ef84c13b266b22732c4c4e5 : FAILURE in 36m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3872589a6c94cceb00bc8007f6bcd71 : FAILURE in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce4594344d7d4fbcb0b82e708292f696 : SUCCESS in 2h 36m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1fc683f647214d5aa8b5de7c365b3170 : TIMED_OUT in 3h 04m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2ae534d80adb40c591a0b619ee8a0962 : SUCCESS in 1h 52m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/14e736c9ec024e8a888b9940a058da3c : SUCCESS in 1h 57m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3bd2877262d44f93aa9aa92c3a5f14ff : SUCCESS in 39m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a87e6f1e41ee480d96b1377c4eb8a182 : SUCCESS in 1h 06m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/878b17569fda40dda34e3839139821f4 : FAILURE in 1h 58m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/18e195dbdd3e4647a5c26c580e84bfc4 : SUCCESS in 51m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ddd8fd82d81740f1989d07ae27ebf136 : SUCCESS in 2h 04m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/aa6dff87307d4f5ca4f27d50e30ca9c6 : SUCCESS in 1h 24m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2f595fbcfac4a788549f4e96576686c : SUCCESS in 1h 46m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ffb7f350a05488dafa198101f15d602 : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":7},{"id":"8232f601e1c093ada1ee6c437b6b17ed9e071dbe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-06-18 11:22:07.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_06_2024_11_05_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"d14e33d85b7675a36fb85c218e78015788e3fc87","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-18 14:07:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/9deafcb/ : SUCCESS in 1h 22m 53s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/62c32191e0dc44deb4f54b89c1b83a56 : DISK_FULL in 7m 30s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"1512b101d1171cd7b768f10cd91df54ac3326cb8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-18 15:31:43.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerStore_iSCSI/c724763/DellEMC_PowerStore_iSCSI/1460 : FAILURE in 1h 33m 56s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerStore_FC/28397cd/DellEMC_PowerStore_FC/1643 : FAILURE in 1h 15m 24s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerStore_NFS/010d84a/DellEMC_PowerStore_NFS/1154 : SUCCESS in 1h 09m 25s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerStore_NVMeTCP/d3e1b2e/DellEMC_PowerStore_NVMeTCP/1007 : FAILURE in 1h 41m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"ad3751345562d8437b7f425a8db6df158fc822fb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-18 15:41:04.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerFlex_3.6/b1e34b3/DellEMC_PowerFlex_3.6/1608 : SUCCESS in 1h 52m 34s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerFlex_4.5/b35487a/DellEMC_PowerFlex_4.5/503 : SUCCESS in 1h 51m 39s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/1cc5683/DellEMC_PowerFlex_4.x_NVMeTCP/454 : FAILURE in 11m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"9fe2bf26dce68a8c1eb2076e84a92b514db513e9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-19 11:24:21.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_Unity_FC/68814f7/DellEMC_Unity_FC/2116 : SUCCESS in 1h 39m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/7/check/DellEMC_Unity_iSCSI/a7264e6/DellEMC_Unity_iSCSI/2132 : SUCCESS in 1h 51m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"111258bb1fe4aa3592f44dee929fedc3c855c12d","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-04 03:24:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"70bec7f0f11ab1a8a8609e5ed2dda46411f013dd","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-04 06:05:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-iscsi/08f7b18 : FAILURE in 2h 15m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-fc/96509a9 : FAILURE in 2h 36m 16s","accounts_in_message":[],"_revision_number":7},{"id":"b9c1b83572c0b1368e32e009a790c3bd96aa676a","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-04 06:18:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7a7c34533b7507f3d86b5efd810a39795780a4e5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-07-04 09:01:30.000000000","message":"Patch Set 7:\n\nBuild failed.","accounts_in_message":[],"_revision_number":7},{"id":"7aed76e9a0c0c95dccbdf7e690d8eac0b15c60ae","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-04 09:30:46.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-iscsi/89978bf : FAILURE in 1h 39m 50s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-fc/b7bf726 : FAILURE in 6m 59s","accounts_in_message":[],"_revision_number":7},{"id":"8b66e0cd140b0d8da141fa2216551b7c3ff328a0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-05 00:32:40.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/7/20240705064522/NEC-V-ISCSI : FAILURE in 2h 00m 23s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/7/20240705064522/NEC-V-FC : FAILURE in 2h 47m 15s","accounts_in_message":[],"_revision_number":7},{"id":"25e129487e7cc6b8ab5eeefa034d7fa06da95962","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-05 02:32:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"66023c45971beb6ec890ed8c0467d3b3ce97bd38","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-05 05:48:38.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-iscsi/fcc947f : FAILURE in 1h 34m 44s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-fc/eb99665 : FAILURE in 3h 11m 14s","accounts_in_message":[],"_revision_number":7},{"id":"3b34b59dc91cf75759f076edfc467d2d08d1ef07","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-07-05 06:49:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/912051-7/Seagate-iSCSI/ : SUCCESS in 2:21:19\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/912051-7/Seagate-FC/ : SUCCESS in 2:14:12","accounts_in_message":[],"_revision_number":7},{"id":"488685f43f172fe04971d310a9c9e81c0eedda73","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-11 06:42:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7d65263ca55a505c3a1bee25dba19e0f86cde256","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-11 09:52:24.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-iscsi/6bc4956 : FAILURE in 1h 27m 12s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-fc/90b7d06 : FAILURE in 2h 21m 20s","accounts_in_message":[],"_revision_number":7},{"id":"19166c617247efae3b0aca5ece11adaff429c632","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-12 09:50:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-iscsi/d1e811d : FAILURE in 2h 37m 03s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/7/check/fujitsu-eternus-dx-fc/c294fc8 : FAILURE in 2h 45m 59s","accounts_in_message":[],"_revision_number":7},{"id":"12fbf2dca4411726d8eac388daa0030fc7ca900e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-05 07:31:39.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b77e08ae1c40a675ea7406db080c6a5e6be16e41","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-05 07:31:50.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/215/","accounts_in_message":[],"_revision_number":8},{"id":"363aeb047c158b14a8512b237baf9004f6eb22c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-05 07:31:52.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/215/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"44b5bb34b688e65c514e177c0a2c8527c5763ee2","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-05 07:33:40.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"64c2c643d7df508843052009f8f95c80d361991b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-05 07:38:38.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9afb4df7441368c47d2c3b498218210b502ce4bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-05 07:38:50.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/216/","accounts_in_message":[],"_revision_number":9},{"id":"a0dc644d3752648066fb32d54376814d5f5ce939","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-05 07:38:52.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/216/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"ff8f81d6782027d349abfbb0cac2fbdb1ae6ec6c","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-05 07:40:31.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"1730a71573c9d2129b35b79852281e1cfd942d3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 07:58:35.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/085f9093309d4f348e66b7b0752d7248\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bf2be5369254dcdb2568dbc30fd89de : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bd26b62f963c4e958d0f00a92c59e62f : FAILURE in 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2fbcc26fb7c37af6f2a99db9b42929308097cbc8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-05 08:13:41.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/de7a9aa5be654f28b21b62d6fe085d9f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/63dd8c3509b043c8892c0682c1e2811a : POST_FAILURE in 22m 45s","accounts_in_message":[],"_revision_number":9},{"id":"f21e2afe95e82c311c3238ed4eab09edd6a2af3d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-05 08:19:07.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 34m 36s","accounts_in_message":[],"_revision_number":9},{"id":"be3891bce4f4f4758a627f80938a27a3c378db69","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-05 08:30:22.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 36m 38s","accounts_in_message":[],"_revision_number":9},{"id":"d5cf015c881a798629eea2960c0a49fc3655226b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-05 09:27:48.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerFlex_3.6/4725431/DellEMC_PowerFlex_3.6/1840 : SUCCESS in 1h 48m 03s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerFlex_4.5/440e419/DellEMC_PowerFlex_4.5/719 : SUCCESS in 1h 44m 57s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/459187f/DellEMC_PowerFlex_4.x_NVMeTCP/683 : FAILURE in 14m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"e9ed21678ec76c1ad489c2579c4e5e5112d242aa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-05 09:34:41.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_Unity_FC/fabdbe8/DellEMC_Unity_FC/2354 : SUCCESS in 1h 42m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_Unity_iSCSI/1ea16ed/DellEMC_Unity_iSCSI/2359 : SUCCESS in 1h 54m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"92acf0ee29c6a570c9fe2aeac8cf4e1833da5f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 09:45:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf8fa100a881470b93c8f23eced2c888\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a15cb066c04b9f928d171619ca795c : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffc06bb4ade940dcb107dffd6ea37df3 : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/766d8d9df5054a01b7794e0326ba2c05 : SUCCESS in 10m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b2290b2f4b343b5a360c24d37aca323 : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e41a6b3b7d4a37b0fb24c2c233a3ca : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/fc0f5f06f8364a7aa39a1ebd9c4ef7fd : SUCCESS in 48m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c6adac4e85c44bea8df37dcef82ceb7 : SUCCESS in 1h 16m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/607584d8954a423fa7b1a2c114531fb1 : SUCCESS in 45m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/060e03a85bef4c24947536adbd7d876f : SUCCESS in 15m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8e3930f049c04824985724044afc2f16 : SUCCESS in 17m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b82ea4d03de6433f9398c8dc16618f14 : SUCCESS in 4m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/74f178b0b60848d7b771894a1450c9ad : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/587c7e6c86054b8c8994cfcf6e1132ac : SUCCESS in 6m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9f750377dbf149419642600e15a01ed9 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/12ae7c9bcb0c410fa704d8c10c413eec : FAILURE in 33m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b26f2ce94ad94d428f69ec8ec68360d0 : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a2964d5a0b6458893a30b7568e1ec9d : SUCCESS in 1h 46m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/67a84ad004ac45cca65ae24bbe39701a : SUCCESS in 2h 00m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5a072cc6ae84594b57982605f4dc6f2 : SUCCESS in 1h 57m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a946168d06d040b0afcb0a2f34ba25a4 : FAILURE in 19m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/55f03e25feb34675b9dfb375c8fb077d : SUCCESS in 29m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/245781eac9a14c929a660a818e20581a : SUCCESS in 1h 05m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e060172673394711ae9e5f481673dca9 : SUCCESS in 1h 20m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/99a579de85034be490f7337ade6a8c36 : SUCCESS in 42m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2151c706c6d3418c9e80ac8926bd090b : SUCCESS in 1h 20m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9a05939bb8474d7382a5ad162576b499 : FAILURE in 7m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f132cbf0b0bf45a78ca9f84586faeba1 : SUCCESS in 1h 27m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7174d7f6415c44309c0a5b8d56eeb9c7 : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":9},{"id":"21b030d4290f64f7400f0677d2db0dc6dc3df169","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-05 09:50:53.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_08_2024_09_45_17\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"745a9a9de4668ab5c2af125a7031a4a43d6bdc47","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-05 10:02:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/46108d3/ : RETRY_LIMIT in 4m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/912051/9/upstream-check/cinder-tempest-plugin-ontap-nfs/2a0fa9f/ : RETRY_LIMIT in 10m 01s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"1ce89814d20667242b4c4bececbc2fafe0956634","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-05 10:22:58.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-iscsi/c73cefa : FAILURE in 2h 39m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-fc/0624cfb : SUCCESS in 2h 28m 34s","accounts_in_message":[],"_revision_number":9},{"id":"039b2bb8b2fb1e043e0cc19a6b6c341d5c60532c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-05 11:08:18.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/842/.","accounts_in_message":[],"_revision_number":9},{"id":"51b21cda7d90fe9752e2b2aa1bb3959b42252253","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-05 11:13:09.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/03b34229700244859e74c3319025839d : SUCCESS in 1h 27m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e3462d62fea441ffaf9bdcb2a85da513 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":9},{"id":"21d95cd6608591a7f491a60e859ae8e1e97224c2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-05 11:36:05.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerStore_iSCSI/02b16fd/DellEMC_PowerStore_iSCSI/1568 : SUCCESS in 1h 32m 22s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerStore_FC/b5e3804/DellEMC_PowerStore_FC/1763 : SUCCESS in 1h 14m 03s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerStore_NFS/a4d4e6d/DellEMC_PowerStore_NFS/1261 : FAILURE in 1h 15m 57s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerStore_NVMeTCP/7593224/DellEMC_PowerStore_NVMeTCP/1120 : SUCCESS in 1h 50m 08s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerStore_NVMeFC/c2aec34/DellEMC_PowerStore_NVMeFC/84 : SUCCESS in 1h 09m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"7adbd49a74511c3c7a22e32296a7fc52aaa0d0b8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-05 15:09:00.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/9/20240805200619/NEC-V-ISCSI : SUCCESS in 4h 00m 54s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/9/20240805200619/NEC-V-FC : SUCCESS in 4h 02m 37s","accounts_in_message":[],"_revision_number":9},{"id":"2deac7402baeae2e5f8df1f30e7a0a15b4f98130","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-05 15:32:51.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_Unity_FC/2714f71/DellEMC_Unity_FC/2356 : SUCCESS in 1h 40m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_Unity_iSCSI/2274400/DellEMC_Unity_iSCSI/2361 : SUCCESS in 1h 51m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"938b5809c5cbdec38da791a5a86aa6ca0cc5d589","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-05 15:35:20.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerFlex_3.6/a28a0a6/DellEMC_PowerFlex_3.6/1842 : FAILURE in 2h 08m 46s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerFlex_4.5/988bf88/DellEMC_PowerFlex_4.5/721 : SUCCESS in 2h 08m 33s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/16f0ab1/DellEMC_PowerFlex_4.x_NVMeTCP/685 : FAILURE in 13m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"111edc4b25df5d86aef038b9b0d1deeb749a6af7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-06 00:23:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/9/20240806011256/NEC-FC : SUCCESS in 5h 45m 42s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/912051/9/20240806011256/NEC-ISCSI : SUCCESS in 8h 10m 07s","accounts_in_message":[],"_revision_number":9},{"id":"26760eccb11585118015a81f3263b4ae98049ec0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-06 02:05:42.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ed492127d0d53137c3ca8c5599a760469800eeec","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-06 04:38:25.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-iscsi/5ca1686 : SUCCESS in 2h 29m 30s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-fc/b334ca2 : FAILURE in 2h 16m 54s","accounts_in_message":[],"_revision_number":9},{"id":"4b2df211f98df4cbd7f2a266804dfa54e27d425c","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-06 05:06:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6a360b2d7e9c9403aa8cb4a7c35ae0a099b9f8e9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-06 10:49:42.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-iscsi/b2f3382 : FAILURE in 3h 51m 11s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-fc/e9be356 : FAILURE in 3h 51m 13s","accounts_in_message":[],"_revision_number":9},{"id":"5b1ed02fe180b4200280976d840230961a14f256","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-07 06:45:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7e2de861c71bc3f87014e3ee995bf153181b558d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-07 10:40:14.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-iscsi/06dc7e9 : SUCCESS in 2h 43m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-fc/08ec7bc : FAILURE in 3h 50m 55s","accounts_in_message":[],"_revision_number":9},{"id":"7dad102fc2814b6b92f41590a1360e449342ff39","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-09 01:20:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f310a3442f6c20cd586213e9729f18b961ac5319","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-09 06:58:16.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-iscsi/2d78ef7 : FAILURE in 3h 50m 31s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/9/check/fujitsu-eternus-dx-fc/9c2c939 : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":9},{"id":"7bce0e54ab53e2e7438efbda1464f54686e7f4c1","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-09 07:43:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e4cee83755361c87d146b283216937a767823a69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-09-02 01:43:54.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"63b31330b37d0a445b0d57e5b86c625df0fdd91e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-02 01:44:07.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/451/","accounts_in_message":[],"_revision_number":10},{"id":"6417cb45a061e2e780ee9b72e89331cd61879aff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-02 01:48:58.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"8ab4aa5d9fdbb26293561b53d68a06ef9b90ccb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 02:03:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9031a5132c9342c18a32d55557382d96\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43e6fb8c70fe4d48a92879e546932190 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc09a80503be4e18afd367820b4c4abf : SUCCESS in 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ae2108a88547a5c735cd576dee57710ac5e397b1","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-02 02:38:42.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912051 : FAILURE in 34m 40s","accounts_in_message":[],"_revision_number":10},{"id":"8c49131c43f1a16ea80234aae34a38005af568f3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-02 02:39:17.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 34m 00s","accounts_in_message":[],"_revision_number":10},{"id":"8eecf8c9cf7a65893d1ba2e04f8dc63ffe9b32fc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-02 02:43:35.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 33m 04s","accounts_in_message":[],"_revision_number":10},{"id":"bba1460a9c497a92867c651f9cf5fffff3565f14","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-02 03:11:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/456/.","accounts_in_message":[],"_revision_number":10},{"id":"405005858bc15a09977177ea36def517c7bd490c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-02 03:15:34.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f0b4147fa9a84f22bb861eb63848e994\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea1fca3e7d3e46fc85c4b0317d2729e9 : FAILURE in 37m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0866a87b9084502aeb9b8eb7ef078bb : SUCCESS in 40m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/671567797cbb4b3ab0779fc274de7251 : FAILURE in 1h 30m 22s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1f54100195e844108f59362587ca96a3 : SUCCESS in 42m 33s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"0f8579301d021ab5ad885468d56703e1fffc3ad3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-02 03:41:50.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/56832b1b7f854c1e8745da3178c0fdc4 : SUCCESS in 1h 02m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4e57565e93654cfe917eaba8fcf45d39 : RETRY_LIMIT in 45m 32s","accounts_in_message":[],"_revision_number":10},{"id":"c2b8685179996b3a3a68aaba1cb8374c13b0f95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 03:47:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54abf5bf12f9427aadb9080bff64f6b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c1feb8965ee48a5b1239aed2c83741a : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b737826c9ee40a3be34f2b1e5075b17 : SUCCESS in 11m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/262c18fa5335416d9635bd7a5212fd87 : SUCCESS in 8m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f3b5e018c6f4d5cb07b5e64f9230ccf : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dbe7757cf74453cbb7cfa0b5fccace0 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/6f4446d8cfca40c493dafb15830462b4 : SUCCESS in 49m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1b01259c98f4fe7a11ea0d9895ef142 : SUCCESS in 1h 28m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31d2f2848e1048acb549161026107cea : SUCCESS in 53m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67253c00f1be41d39a2e4b9b9528562b : SUCCESS in 19m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a63f0fc88afc4afc86e733c83f87f00c : FAILURE in 13m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ffd5c0a3407141a59f2fef325b32102c : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3956883dee640ce881585d1052d7c9a : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3ebcd83fe7ef4e62ad56747f437dca78 : SUCCESS in 5m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a6fb55e80c0d4308ba1e312883276d7f : SUCCESS in 5m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/08b4465acdae4ca9b6658bb30fd795b8 : FAILURE in 39m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85fac725fd46410d87baa39721f99fda : SUCCESS in 5m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3e73d6c42b5e490dbde79017c5d66164 : SUCCESS in 1h 55m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e289218d7c8f4a7397f6e55e9848a95d : FAILURE in 2h 02m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0e82ce70882a4f3cb28f65e8d6b3ebd9 : FAILURE in 1h 49m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f4b0b5a87bac4bfab681b9af788fc861 : FAILURE in 33m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/12aaf837b1db4413a0383a4bebd67af2 : SUCCESS in 40m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1a838587eff54e95abbd73dc3a98e2cd : SUCCESS in 1h 03m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9382dabf5ec34aa3b77ea32324f7ab1c : SUCCESS in 1h 22m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b41e15ae4eaf45419d1f817cc6493066 : SUCCESS in 28m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/44ccc3f4ec6d4230bb63f164b4d01180 : SUCCESS in 1h 20m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e5bbd9eab1c84df1ac97b9b3c6a2f3f9 : FAILURE in 15m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b69ce37abb4142ecb726034ec98339bb : SUCCESS in 1h 21m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c23cd94ac9ff43c791f2d7b9802db9f9 : SUCCESS in 43m 51s","accounts_in_message":[],"_revision_number":10},{"id":"5c3fe1611bc7905d53dde2ac2af94ea1d5eab91e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-02 03:49:00.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/10/check/DellEMC_Unity_FC/d57a9c5/DellEMC_Unity_FC/2524 : SUCCESS in 2h 02m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/10/check/DellEMC_Unity_iSCSI/742d004/DellEMC_Unity_iSCSI/2529 : SUCCESS in 1h 51m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"bc888e1ae06fe17e213ebb20bc1276d5cb0f9b6a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-02 04:08:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1044/.","accounts_in_message":[],"_revision_number":10},{"id":"fd88d660cf500b665fe1ca82e9bc112fee37718f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-09-02 04:50:50.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/10/check/fujitsu-eternus-dx-iscsi/68665aa : SUCCESS in 3h 00m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/10/check/fujitsu-eternus-dx-fc/2dc9a0d : FAILURE in 3h 03m 47s","accounts_in_message":[],"_revision_number":10},{"id":"a5c22ac63f3084222f5194f81e03d560674f705e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-02 15:39:43.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/10/20240902155635/NEC-V-ISCSI : FAILURE in 7h 27m 57s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/10/20240902155635/NEC-V-FC : FAILURE in 8h 43m 03s","accounts_in_message":[],"_revision_number":10},{"id":"a5ca8bbd9f9c7d37d40fb40caf6b97d047113201","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-03 05:25:08.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/10/20240903074853/NEC-FC : SUCCESS in 6h 22m 00s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/10/20240903074853/NEC-ISCSI : SUCCESS in 6h 36m 10s","accounts_in_message":[],"_revision_number":10},{"id":"c22b3ab0b46624c83bc7f445b2e7102f5049cdee","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-09-06 09:14:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fe0d06b5a0b5d10dbb5fddf11c4ef84fd15279f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 10:02:31.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5bd10595b2493fa1d501b02697c81c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e18750ef6ec6440ab9c64ae7c0f6b237 : FAILURE in 20m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18f835e674474cc98d805ff3d55cc6d6 : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f623492293ee2d14f6233a4f08af01f3a4ff0a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 11:43:17.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d80180e9d3640318da4cb3744129bff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b330337d0c94bb9951ab7a251f9c346 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1687216705db48fcb5844efc5db9e57d : SUCCESS in 12m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf42eb37de374cad84a6df74875721f1 : SUCCESS in 10m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24adaea2c18c49b3b111935d98ed5b83 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/953ff2fe13244b34aca40aff2ccaa4a4 : SUCCESS in 16m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/d8a4cadcace2435d9aabf229fab2262b : SUCCESS in 1h 06m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29a081088b014374b9310a9e7a9a6781 : SUCCESS in 1h 43m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b7cbeee464c41a1900590c9262b19e5 : SUCCESS in 1h 01m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbbd52f4f2e4440fa5f1a97a4695f3a0 : SUCCESS in 18m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d64d18bf160f429db3076ef09fb71beb : SUCCESS in 16m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/78caffbae415443fb1b92df9ca171fb0 : SUCCESS in 8m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d0fb5b81a99241939c263d5647579c10 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6977488e58e447cc82915775694ed652 : SUCCESS in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dd5217df084a439bba2fa78b1ba15630 : SUCCESS in 9m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2809e70bfd8b452594a08d3a4554ee82 : FAILURE in 38m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/532a5c10b3f94ea49c21058f4cd435a3 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a32b6f4699a4ac2a0bd188e1e378d16 : SUCCESS in 1h 47m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e098408390ea4e9797d6070c93e09127 : FAILURE in 2h 20m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f155d947959b4e168be8da096ac1465e : FAILURE in 1h 42m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8b635c36d3a1405fbb622a98c5965512 : SUCCESS in 1h 59m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/307f5e7494844228aa7d9c16bd536b66 : SUCCESS in 37m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4c1170ad96a483eac2acaed872e16b8 : SUCCESS in 1h 09m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/72f2b20eb595462bbd2102926bb5d35e : SUCCESS in 1h 30m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76ba84078d584be4bf132cedc8790070 : SUCCESS in 37m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3c1cacc596df4d80a3a03849edb4d634 : SUCCESS in 1h 39m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/902dde8fde2c41d6bbf6997c40f5dcb8 : SUCCESS in 1h 46m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e408ec0a47f4c008b7340ed8b7c66dd : SUCCESS in 1h 37m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77063b6849c341198d181a3fbf0d6b46 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":10},{"id":"7120d7940f37994732d8a861a7a1d4ef1a6df8e6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-06 13:13:46.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c410f2507817473b9278602e668f1d03 : SUCCESS in 1h 10m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/354c8e85af0c4b4ab3b8e5811f3be8d1 : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":10},{"id":"7800a97e513f90717c5dab00bef5af4daca1c995","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-09 08:16:48.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b862a3a17ef11afd5d7e396fd693c77bd95bd63c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 08:37:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b39b52ecef4a3985654259580ac9cb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f416684cb374c11b6ea0c7da691482d : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d365336ada58456ea497d34108f6ed7d : FAILURE in 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c59945d74936c96d7404cfeafd50790ff58e4405","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-09 09:52:04.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/50616b6a4f9b481cac7a8f6b8d73f201 : SUCCESS in 1h 17m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0892cbba805a4b33a650d1c02c5d56ac : FAILURE in 1h 32m 16s","accounts_in_message":[],"_revision_number":10},{"id":"3466b9cbd70ddbdce630f2b016e2ee752cface46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 10:50:06.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55138070bf0747a185bb0878c042ae47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40368ab8684f4c69b2342d28ee4bd401 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e992aee6607c416fb05eab39b6374afc : SUCCESS in 12m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d6cc62c74854653a12f06fd442086cc : FAILURE in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a23e329f5f494385b589d764d79fa6 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/bf77f6656b7b42b998f018120aab9cf1 : SUCCESS in 56m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2eb1a51596f44238a002deca623494df : SUCCESS in 48m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd4b451117204abba65de48e09828cc4 : SUCCESS in 1h 26m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ed92e0839c944139d7b2e8bb7058cfa : SUCCESS in 59m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/605d53d749094ac1b087348e2cfc06d4 : SUCCESS in 19m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6c2612a4dbe74e10bc39b0b16765130a : SUCCESS in 18m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f6f50d70575f446488176f82dfbaea01 : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7dd8e6712d334df0963c489c4ac8ced2 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38962837a1fb4adda324eef636f73617 : SUCCESS in 6m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4db80a266b9c43cfa2c18fa6eb43db88 : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5cec47e14d1b47f784d51d1a028fad49 : FAILURE in 33m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6299da3b6c0b4b7f886bb590fd44d3c5 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb547293358b4f6eacec949762e1fa54 : SUCCESS in 1h 39m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b5f42ca5f72f4fb18beef21c85228d24 : SUCCESS in 2h 24m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c505382a634d430abaf4c8743b674591 : SUCCESS in 2h 01m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ddab4e71e7004d98be215188fdc242e8 : SUCCESS in 2h 14m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/61a686098eb5421f87dbeacbb0f73d05 : SUCCESS in 33m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd1ba9999ab74974aaf44cfe83675753 : SUCCESS in 56m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1c9c24465fe64eb7897ca76b95220bd0 : SUCCESS in 1h 38m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/eb34c91043e24d9fb7c05a5d1b8acdba : SUCCESS in 39m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6c1d0d1626bb4b458d0f1e8ff42e43e9 : SUCCESS in 1h 45m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e4c24c6228aa416c84840c4205f0adf9 : SUCCESS in 1h 30m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f84767e67224e6793b4ec018f7c6db2 : SUCCESS in 1h 12m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da5c6c0467c34787aa26fa26eb5b5d77 : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":10},{"id":"b7fabba207819c0a059574abc466bed46c2b5005","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-10 02:55:21.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0f46e89809ce1b442245b3c7b8e3e0c8b2b1be8f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-10 03:11:21.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e2a5a59cfa4d4a4392650bfa0cb52e17 : RETRY_LIMIT in 3m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/55eddcec880544bd87b9e7f5ec2b6885 : RETRY_LIMIT in 4m 38s","accounts_in_message":[],"_revision_number":10},{"id":"6914bc98ac992075c85684c1f5d7f8f983c918a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 05:26:32.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81b68cdfc1f34f32971c98afe7b815b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae1199504289473f8e32003a444fe00b : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe7dfca8cab641b1a502d61b679525ce : SUCCESS in 10m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ed3b9583d1f4e1b8e0fd07ba5c904f0 : FAILURE in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0201ef5123924c688ee6ad8ced1e5cbd : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/f7b48425e75a4eb9b95ab4e6bece1c69 : SUCCESS in 1h 08m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/06299c7c80c445608633e2d9611b1434 : SUCCESS in 1h 01m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5940b74e1d52432e96353c14187c16f3 : SUCCESS in 1h 50m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73e4acabf68a4752892f60d01eb4be49 : SUCCESS in 59m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50b6b872e55e4b9a9a232f0a6d66a8b5 : SUCCESS in 19m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/76434f06ce2d49ba9dad41a172ffa9fd : FAILURE in 18m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ed73aad64c394211a46338f719db58f0 : RETRY_LIMIT in 1m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/caf458922b1d40c4989bdd21dcdadd6b : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bd1f3101cea042b3866a6080fffc2a7f : SUCCESS in 10m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4c574471a5ef4219853a108f06f66d4c : SUCCESS in 10m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7678e37dfdd24479913c122e1843da10 : FAILURE in 28m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a71ae87f52c145c69a569799d44f28f3 : SUCCESS in 10m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4d156b66c2b54e63a8cd9cc70eac4a24 : SUCCESS in 1h 41m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f66526bfdd804e1ebc9592ae5942a2b9 : RETRY_LIMIT in 1m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3334dc3ddabf4d72a4276c93bfaac06d : RETRY_LIMIT in 2m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/24bedfbaf8cc40dda64bfb340ea09df5 : SUCCESS in 2h 09m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dd1f2454414443f29e7685dcc0ead8a1 : SUCCESS in 25m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7e76438f05744142a9e360e4270280d8 : SUCCESS in 1h 19m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d7f6ea599cba4a2982c65f9d44e91c98 : SUCCESS in 1h 34m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4e04e03dc7164fb9b10264db9035a7ca : SUCCESS in 41m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ea4055e36be34c11a64201aa1d95ef71 : POST_FAILURE in 1h 39m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1c715c1a4a96455f9499a6676756b01e : SUCCESS in 1h 46m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd6157ca55084047807d27bb3b57527c : SUCCESS in 1h 34m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb2332dfbb0d4e718199dd717f803a80 : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":10},{"id":"64481243ca7c54fedb1f96e6c53b7f153a653da3","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-10-10 12:00:10.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/10/fc : SUCCESS in 1h 00m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/10/iscsi : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":10},{"id":"37ad82b1a084c4a5f678540c71aaf4ed126e323f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:37:37.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2d379650ab433ebd0259feef93161d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/755c567814e44c8cb4e797a9bb0b29e9 : FAILURE in 24m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f68db48fdbd48018b646df6c941bde0 : FAILURE in 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bd0baa8ff624b108826f292e5d3d159e42a5b516","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-14 09:40:21.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4d5f46362b2ba43430577662075ada77dbd3a6e6","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-14 09:42:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4fc7d0689f9bcbd7818fb84b78157ff414394758","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-14 09:42:34.000000000","message":"Patch Set 10:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":10},{"id":"a10caa11b81086ff464eb4049d84e05067bb5f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 10:13:21.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc396a46ffb24d528577fffd5a414ef1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7a55b011d1d427fad6616ff9ce432af : SUCCESS in 28m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7e4e33fe546444fb129ead3ce9e1879 : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"775c25ba32e4f92afb2be58e8e3bf732adbe5990","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-14 11:09:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/06f7c78b80d843778d7b597e61a7e357 : SUCCESS in 1h 19m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7a99f9c44219404cb39e508bd1c2fe6d : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":10},{"id":"971ddc3eeb5566bc68575ae9ebfa30c75bdab130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 11:57:28.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0187d8159cf74454bc8e289f4631a39c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/346190f647c4431b9cfd8be5dd85a4af : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8a8a5cb67574fd2895e00563c1f46e2 : FAILURE in 12m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0860d65965a49deb530a2ffb03493dd : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffd044f2b9c49a39f03429ae036066b : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/88a9c4b13f1644c1b668ae5ff0637b86 : SUCCESS in 53m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/994187cdc95d499b959941436bfae5ea : SUCCESS in 1h 11m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c133f95ac0ed4ce697b62d665bc291bf : SUCCESS in 1h 34m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc2ce299d3e04de9b80f49de4a793041 : SUCCESS in 1h 06m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4401feb83f7d4108acac85731128ca42 : SUCCESS in 21m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c48c6a07a3094ef29e53348d8ee7f648 : SUCCESS in 17m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/86747c0b7a244716936e9e96bf74bea7 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8185350af574229a63279c8d5e7eee6 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/574f61a1aef04d6e9411c006bc5f8e15 : SUCCESS in 5m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f8666258d8c04d8b9ab8510d91de8c86 : SUCCESS in 7m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/39944bae19ef423b867898c552dda200 : FAILURE in 19m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b609729fdf204aef97c8f6860b2596a2 : SUCCESS in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ef947a068aa4fd7beab2d48bf5874a9 : SUCCESS in 1h 41m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/db64980723ec4e5dbc3e9c6ea39d2c01 : FAILURE in 2h 03m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3df0fb2e45ff4734ae64fd6f6a3fdb7b : SUCCESS in 1h 43m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ac751211bb6a4b2b89ebdf483e120c3f : SUCCESS in 2h 05m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f908e56cc5dd43c49f24742e2549278e : SUCCESS in 24m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c1589aeed3fa40f4ba9fe12114bd373d : SUCCESS in 48m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a991e1fe73943a3a36e4d55ef5e684a : SUCCESS in 1h 32m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54963a8cc0c04fe59165bd09136e53e6 : SUCCESS in 27m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b3cc3d59b1a24d68abc530992e3c5190 : SUCCESS in 1h 44m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7d2cbc1dbabd4967bb898ce0b696062e : SUCCESS in 1h 24m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/157c0697b77e4f75b4eec5850614db40 : SUCCESS in 1h 29m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c50146e1e19478ba3c4e115a040ef15 : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":10},{"id":"6274fad67fd69e20a6f4b7c6a8c7de692602071e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-14 12:07:33.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/10/check/fujitsu-eternus-dx-iscsi/c5ecaa2 : FAILURE in 2h 20m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/10/check/fujitsu-eternus-dx-fc/15d69fb : SUCCESS in 2h 23m 49s","accounts_in_message":[],"_revision_number":10},{"id":"f37070889d2c38657c61986a644e8a7d6931f126","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-15 01:18:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e9379d1eb9e73e024075a99c1cd0abf3ffc19da7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 01:46:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5058ed6055d340de813ad8c6770c22a6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69d0328cba724e15b5e75ab3c5e0692f : FAILURE in 24m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3765ea8bcfd42cb8d053d0726f292c9 : FAILURE in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bb496e3e81abaaab4308e8b3e3d59228b54b22f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 03:24:52.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8b20ad4f54d417183cd7fc6e9a073cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a647ef5960e4ae1ba8e545ce767146b : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e77edca9f584b6198df17783378cd3a : FAILURE in 10m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3a91184fdc74f4cb251cac9b01a6455 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0329efc56b54818b022a019e56ff1bf : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/43fb9e0d03214b349637da47672dd6a4 : SUCCESS in 1h 05m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e31f189bde93438e9ea27ebc9c790661 : SUCCESS in 41m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b0b97e54a3649a9b2a61f42578f6a2c : SUCCESS in 1h 24m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10c1eac48d6644ecab46de0d7c58a607 : FAILURE in 38m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e10822e351564e50b430ce54c8036aa7 : SUCCESS in 22m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d330fe69acd49b8adfddf6684876ff4 : SUCCESS in 14m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/db7849cf4a6843f0a2cfed47914a3088 : FAILURE in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6eb977dcca2847368882c1f20203202f : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2318279d7dd49f98e459641386c0842 : SUCCESS in 6m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7b5469baba76492cb97d87e66ac0738b : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/29be3e76414e4fdf9f0ce80ade9aaf86 : FAILURE in 17m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a67131b11cde4685a8fc5c635df80019 : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bfb1fd3000d040ec8f557978caa2d998 : SUCCESS in 1h 47m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1d1ac1856e424a5b9465e7fc8bb13a0e : SUCCESS in 1h 39m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4fe11d80938a4180a039a6658cba9acc : SUCCESS in 1h 55m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/53d63d4ca6ea46eea5196ace4aae0d3a : SUCCESS in 1h 46m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/517434fbcb254c6a8a6104aa80f7d03c : SUCCESS in 38m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ceab15a047984bbebbb506238957ab57 : SUCCESS in 43m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c91e2b3c80e24837a18cc079174cf947 : SUCCESS in 1h 24m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8d398ec98f174e289660054cf046f16e : SUCCESS in 40m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ec7d02675ff544a1b9ff0a0784f1daab : SUCCESS in 1h 31m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4edf4c2230be48a19a7c57aa45173e7b : POST_FAILURE in 1h 47m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fcd9eede8cf4db1b1024ec929899b65 : SUCCESS in 1h 52m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/390dfe85064a4c38a41af46027ff34b7 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":10},{"id":"572ec71433c984aeadc3f658d015a20a77807de0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-15 12:31:18.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/565fba371d854201a13601cf2ad0f8c8 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/543a15b98b634498bfc60d90f2fd96c6 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"359b46417af4303327cce38b399716bd0ab6e35e","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-21 02:45:43.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0333519608c9fbc3c026122745a387b37135ae64","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-21 03:10:33.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"db4d148202ecc4162d21fa4d296cfca53183e935","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 03:12:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898f75fc64a2474594db57bbdd750de8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18a8a152407143a1b185c1f2e7ed8b9a : FAILURE in 18m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c2d80255f4743a88524797aa4dfeb62 : FAILURE in 22m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"09c28b06b7e85e5336ed627d84ab04e295b71abf","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-21 03:27:11.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/884ff8ea2d0048d98243de55aeed8322\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/746729c460784818a5fa9f46d80b168f : NODE_FAILURE Node request 200-0000000093 failed in 0s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/a776692156034170ae0e09c5dd2b3810 : NODE_FAILURE Node request 200-0000000094 failed in 0s","accounts_in_message":[],"_revision_number":10},{"id":"41c1558a5a2c27e5acb85a98e99a5fc7d2957af5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-21 04:05:58.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/816733919e964c83b99a4625bac52fb4 : FAILURE in 1h 11m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9da93445467e4ab5aaa7e18cfffc5874 : SUCCESS in 1h 17m 05s","accounts_in_message":[],"_revision_number":10},{"id":"b39bfecee75220199a75384f656d8652ed057d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 05:13:24.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a34f73cd06540789b6f0c310677efa6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec19f589312463d9d485554b695c489 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edfb4190120d430c8c26927e95285d6b : FAILURE in 12m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f92b5f936f245f8b722048caaa48c43 : SUCCESS in 12m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa954dde91e149d6913b9bdadebca5fe : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/e85b43c52a114ba599ac85c5f96c7ed8 : SUCCESS in 44m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ef9075b836d4c80aa4857e7f198b09a : SUCCESS in 47m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4497a03b7a404c7db707649075cb4179 : SUCCESS in 1h 41m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e60f36e92d764a2ba2f6dd153e430d47 : SUCCESS in 55m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afec1d0814494b478ddf17025a97dad3 : SUCCESS in 14m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/108164e22c0c48d8a31478c3b728b6fd : SUCCESS in 20m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8f34e9c89b8541cb9ecb2e05356f7d29 : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f6d6574e7544c5fa1766b4682b14d35 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7413da629df43e2b13c6f829d6c7162 : SUCCESS in 7m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b2dc03063c5a46a19b1e9183aeb51918 : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b2bbba0adb4948a9abe2ce23fccdb0b7 : FAILURE in 21m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cf84deb4af86403e95ba8e0ab41ba26b : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cea7faee52f7414d88007e544b19a3b6 : SUCCESS in 1h 47m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/92e5ac3256c942359c23a428f68fde33 : SUCCESS in 2h 18m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1bbcb47e6e6f4ea998ff5730f3ec290a : SUCCESS in 1h 22m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2a09627573bf4527be5c41cd2c6a6052 : SUCCESS in 1h 25m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/99f9f0e676194af3966c7893aa3e858b : SUCCESS in 27m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a73dfcc8ca04d529fccd20747579f53 : SUCCESS in 54m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bab97abc41be48bb91e63f0f029ea51a : SUCCESS in 1h 18m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/699267799abc49bc9bc12106e74ec04b : SUCCESS in 33m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b7954df6ce8248a6ac19d723980c0157 : SUCCESS in 1h 20m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c304d329e596428c8f86385781548a37 : SUCCESS in 1h 28m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd33607222cd4f6db11d69ef0a57857e : SUCCESS in 1h 35m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e90dc7a5f5648f29dd803a78d11320d : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":10},{"id":"778b36d9a7f55118d73527081c1aa51449259ac0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-21 06:19:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6d9e74d4604770dab9c059481a5cb518eec5c800","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-21 06:21:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ff22921c7334325b11f7a1869bcf0c3c9aef7c9c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-21 06:37:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/d5895f43512747a8a5a421851649a250\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/f0a4eee4c068494986be20048a436f7d : NODE_FAILURE Node request 200-0000000237 failed in 0s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/353bb5f4c09c491187f6ced1f884683f : NODE_FAILURE Node request 200-0000000238 failed in 0s","accounts_in_message":[],"_revision_number":10},{"id":"5f6eaa32376e331b05a86ae9ee0fff90f8256588","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-21 08:45:49.000000000","message":"Patch Set 10:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/10/check/fujitsu-eternus-dx-iscsi/aba7a80 : SUCCESS in 2h 23m 12s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/10/check/fujitsu-eternus-dx-fc/cb2db96 : SUCCESS in 2h 19m 23s","accounts_in_message":[],"_revision_number":10},{"id":"dcbd86601d12820c98cec268621e570fe83e5b96","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-10-29 06:38:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bea548ac5385535dde5460376183361dddfe39a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 07:06:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d26e487429af4e749c78d4ff027782ae\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2291f31940c1445f90c93f9f75181fda : FAILURE in 25m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7402529fab34d598594872894703050 : FAILURE in 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"666f6605f2a12c7e3beeb224594ef95d2e0866b0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-29 09:12:28.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a223a96cfb454ca6a525bef36de8589a : SUCCESS in 1h 13m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aea83effdcd74b5791a5d75fbe49a2e0 : FAILURE in 1h 23m 18s","accounts_in_message":[],"_revision_number":10},{"id":"205772e922085da113416a656669e48843d9dc35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 09:31:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5074195a26494892a1329054eee0d2ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e190dc0a8f9e42628d2f0d7f4805087d : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47f7c427f0e745e8bd68fffcaaebeb88 : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15bb0661a3dd419fb54568457896d464 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d267e7acf41f4ce2b3d8b71633fa3a84 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/e97a7cf659df47f698b116a636708ad9 : SUCCESS in 1h 03m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/841a55d7ee9c46c7a588da3e421228ac : SUCCESS in 58m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a91c331816274fcbaf8e6e71abbbb572 : SUCCESS in 1h 28m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65f1a80c22dd471399c9a83413900461 : SUCCESS in 1h 03m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a55ae2459d947aaac66ddef1565e02f : SUCCESS in 17m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b9382c24a106429995da8c47b64fa8da : SUCCESS in 18m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8dca02755c3c45c3ac0bdc54b35edef0 : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ec3f5cea47ff4fceb8f087f093b41a5f : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e50fc8cda2f48fda191025d5a712d3c : SUCCESS in 5m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/93f77b7cfd4f4a20ba5e684d2db809f4 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/61a292cad37f4401a4b49283877a931e : FAILURE in 30m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/810e603b393e4dcc85fdc285c9864ad0 : SUCCESS in 4m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2d1246543095443f89e1dbf4dbcce359 : SUCCESS in 1h 59m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dce9cb93adba4506894732ec70dba898 : SUCCESS in 2h 38m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/204d0d213dd049f58b51ac5cb1783e73 : SUCCESS in 2h 08m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5ffa45f7b1a448d3b2f3cb5148c2511c : SUCCESS in 2h 07m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f789ebc0d6514e4fbda2b5f8b5d0e56c : SUCCESS in 23m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/071fa6f5fb6d42e59604859503ab5a19 : SUCCESS in 45m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e859356c9b6c4c4e9f701911e98a048d : FAILURE in 1h 44m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/021cbc3520484a34bcd126231b465cc1 : FAILURE in 43m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0bb313e043e74bb593cff12e508002bc : SUCCESS in 1h 02m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c6230b50f82441cfab92eb005ee713d9 : SUCCESS in 1h 43m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/164f47ba1fbc47f5aeaf152a4099848f : SUCCESS in 1h 10m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e147bb658d34c0c9de3c5fffdd1a087 : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":10},{"id":"0e45c2626f3f94ab26181d20caa77c1fbacbb1de","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-29 09:55:20.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/379fe5c804aa4e599b1d41c20cdb5f5d : RETRY_LIMIT in 5m 46s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"ecd9b11b699a7d4366f425c437eaf5e0dc263de6","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-29 14:49:03.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/912051-10/Seagate-iSCSI/ : SUCCESS in 2:23:31\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/912051-10/Seagate-FC/ : SUCCESS in 2:17:01","accounts_in_message":[],"_revision_number":10},{"id":"ba6c2b746f6208ec63f5bc2cf61f69f26dc29cf7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-30 08:08:53.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/10/check/DellEMC_PowerFlex_3.6/6e0890d/DellEMC_PowerFlex_3.6/None : RETRY_LIMIT\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/51/912051/10/check/DellEMC_PowerFlex_4.x/3151566/DellEMC_PowerFlex_4.x/1197 : SUCCESS in 1h 43m 52s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/3109f1e/DellEMC_PowerFlex_4.x_NVMeTCP/1108 : FAILURE in 5h 54m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"7331b97c4a0eb9556bae111866c27c8d805448d7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-30 14:09:45.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/912051/10/check/DellEMC_Unity_FC/35381fd/DellEMC_Unity_FC/2809 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/10/check/DellEMC_Unity_iSCSI/b36a1c5/DellEMC_Unity_iSCSI/2792 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"a9699be951761b524702383b7d46f8b3badbd5a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-12-13 07:18:32.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"b6928fa2e12b7135b5113e693d80a79bf3a76ee2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-13 07:18:39.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1092/","accounts_in_message":[],"_revision_number":11},{"id":"20573e8d70ebe42bedaeacae671d8849d176a0e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-13 07:19:33.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"b2ae8c3d56c089806d13388cfe0efe1114aa0702","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 07:42:24.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feb6b9e586c24a98ae8832497d8ec177\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2f18a025b98481a9cde0066a1c2bdab : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ff55bdc58114006b2a929c385488ccd : FAILURE in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d66a0f811b7f00a38cc9c7220ccc5f7091d3406e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 09:12:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1436c127378b49b4968adbb94d0fa7de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e3b3547d2a541c0a6c90a287a244978 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ee896e89d094a4bb031588337e8bb3b : FAILURE in 10m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/394435e8d43a4f31bcceee814d9916bb : FAILURE in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f88edd3122ac47e0a71f0164f940a546 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/0c612e2d4a9242b081be9815acb6fd34 : SUCCESS in 37m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a8262dffa7b49d1b4e7dd0e8a7ec6b9 : SUCCESS in 51m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa90d2b666dc438aa62c90d77df392aa : SUCCESS in 1h 30m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c856cc4e3bbd4e2b8511f8ec85a5f936 : SUCCESS in 1h 03m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/772cbb199bd3413e96f451e640de4a65 : SUCCESS in 14m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b1a650e80121405ab8641669cc57550d : FAILURE in 20m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4f6b2ed8e01b42e2b97185ce2a43005a : SUCCESS in 5m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f2a44133f324fa5a597bd8b2bc65d05 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb9983511a1648a9b291d87ef0543e3a : SUCCESS in 6m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/01c9f7e107fa483fb5f4480122dc7ebd : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8512df262a3e401082b1b26883f68b8f : FAILURE in 28m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/09a9a7b9f5db495d92b76cc75ad259fc : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/036ed6a6fa924471896475b814c01dee : SUCCESS in 1h 51m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4c99a4035aec42638f7ed8f4f154d302 : FAILURE in 1h 35m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3fdd6751eacc4d289f7806a67ee29f5d : SUCCESS in 1h 30m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/50afddb9958846d99ce809577db4f80d : SUCCESS in 1h 51m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/85df5afcd6944621be34481fc45a7d44 : FAILURE in 26m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5620a0ee53cb4368ba10d2605d9ec9fe : SUCCESS in 1h 09m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2b6748a27730443484e9c926198d41da : SUCCESS in 1h 34m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d3516613840444e19a47ea3b30efc3e3 : SUCCESS in 33m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8e32c44106ce4867bcbc383ff630e7f6 : SUCCESS in 1h 22m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/748890b283754a5fbcf36a3db6a9a455 : SUCCESS in 1h 23m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c8d4de63ae54f6ea9bb8a6db086d684 : SUCCESS in 1h 36m 19s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af5f972860a844eca205e47a8ac704e3 : SUCCESS in 1h 06m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac930e8217674b03b2e909f1b265e7e6 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":11},{"id":"4efed71d00deb2c0e2c003f68a654b235529335b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-12-13 09:48:15.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912051 : FAILURE in 31m 49s","accounts_in_message":[],"_revision_number":11},{"id":"ceb717220dd89d78a85454321a7997e95cbc0bce","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-13 09:51:20.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/66e4d98fbbb6456087aa4c49220e14a0\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6f1512c20704dfc92b6c2ca364f10cf : SUCCESS in 38m 42s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/74c2d780c8de46a19752f2dab6ba3b2f : SUCCESS in 38m 39s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7fa628510d24eaa8269f9fbd2190e33 : FAILURE in 2h 08m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a59d6b520f674b2ab429cad1fd9db6d7 : SUCCESS in 36m 11s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2841f77da2b840f8ac64a72316edd169 : RETRY_LIMIT in 37s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3cf830a611fe4bd3a9f63554975bcb3f : RETRY_LIMIT in 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7aae58ff6ed44a7da2717d6676f1417c : RETRY_LIMIT in 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/28dab266075e4559ba2391654cb3fe43 : RETRY_LIMIT in 37s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"dacf70501687280502739c56315a8f8b2c12b9c8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-12-13 10:33:32.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 34m 25s","accounts_in_message":[],"_revision_number":11},{"id":"e6f27fc879cd712abf84ff118bce41d35d267beb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-12-13 11:10:00.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 35m 01s","accounts_in_message":[],"_revision_number":11},{"id":"ddb7d408210e05a63204f141ef39eb2fcc0813c3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-13 11:24:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/110/.","accounts_in_message":[],"_revision_number":11},{"id":"41c32e8befc3c48cd778018420c16fad8ec80167","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-13 15:30:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/109/.","accounts_in_message":[],"_revision_number":11},{"id":"5e7517011e4213fe92f8b64f76c27f369330aeb0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-13 15:45:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/96c930b292ad431eb006bd771d9085ac : SUCCESS in 1h 13m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/16cc8ce7b8a847ff9dbf342472f1edf6 : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":11},{"id":"ffde642fae3dc9518baa4f351506a515fb94f816","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-12-14 14:36:48.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/bfd2c89eaffc46638128ba183ed91db6\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/2bab7d00920e406791f7c0bf326fed9a : RETRY_LIMIT in 11m 32s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/f0bfc93b7fe84aa6b6aea60dd662a6ec : FAILURE in 1h 49m 55s","accounts_in_message":[],"_revision_number":11},{"id":"813633ba4a7d80be1eb2b92325110b02c62be889","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-14 17:16:19.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/912051/11/check/DellEMC_PowerFlex_3.6/41ab10c/DellEMC_PowerFlex_3.6/2579 : FAILURE in 2h 49m 44s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/51/912051/11/check/DellEMC_PowerFlex_4.x/e882029/DellEMC_PowerFlex_4.x/1418 : FAILURE in 2h 01m 54s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/11/check/DellEMC_PowerFlex_4.x_NVMeTCP/002ae23/DellEMC_PowerFlex_4.x_NVMeTCP/1294 : FAILURE in 4h 33m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"65844067e969a0a5a3afd261235239a585b03df1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-12-17 21:10:37.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/11/20241218021705/NEC-V-ISCSI : SUCCESS in 3h 40m 30s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/11/20241218021705/NEC-V-FC : SUCCESS in 3h 53m 29s","accounts_in_message":[],"_revision_number":11},{"id":"9fa5f9f8faef1a389f1488ca182a64c68f438fbd","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-07 06:00:10.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bc33dd78f0a642ace2fca356f7fa8cfad52cbda7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-07 06:01:19.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 910495,12","accounts_in_message":[],"_revision_number":11},{"id":"15b5f8542770673a6f2505fbdd68cd633abbb742","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-07 06:10:34.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"70a10f741288032284bc55beb9068025fc1848a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 06:10:44.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1150/","accounts_in_message":[],"_revision_number":12},{"id":"9c295a7a4299db6be650b7f28a25f6811e94801b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 06:11:38.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"daeab68e64dd83e9978140d689a35f29e3094673","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 06:21:38.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/188/.","accounts_in_message":[],"_revision_number":12},{"id":"8a11e74d04848d412430fbc408d545a76675910b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 06:34:35.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/197ecc0c725d46c9bb32036f2dddc183\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f16a961ae8b467c8523984e26f035d6 : FAILURE in 21m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c01bbd8aee144a096efe0eb4afceb3d : FAILURE in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7812d963885738d9c70b975737d14c8272989f2e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-07 06:48:17.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912051 : FAILURE in 33m 42s","accounts_in_message":[],"_revision_number":12},{"id":"f500a3015106a69fee7533859944670109cb385d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-07 06:50:02.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 35m 12s","accounts_in_message":[],"_revision_number":12},{"id":"0d98c50f0d9b45753644b108796a389738ae5cd8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-07 06:50:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 36m 20s","accounts_in_message":[],"_revision_number":12},{"id":"2b5c144b493d6382576526928d4a6c56dc9a63b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 07:27:54.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/177/.","accounts_in_message":[],"_revision_number":12},{"id":"b091b18c932005d7931b835ba3ee07e76d5953c6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-07 07:37:26.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3cd73705ff7f4638830e6b2c8a1b4a86 : SUCCESS in 1h 18m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/85d305532a39466c85a6a4fab7e48e2e : SUCCESS in 1h 23m 39s","accounts_in_message":[],"_revision_number":12},{"id":"033ead4c9c099e54279d31f393a65abad8c63ebe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-07 07:46:23.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ebad7218e72d470abf9e4680d76aec24\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1191174f809b4229afc4d91caafe52fe : SUCCESS in 38m 53s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ac8c0b58412f4c0faa2925853c3bca51 : SUCCESS in 42m 06s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c800d3d76ee4ffbabd9479ca5f5edff : FAILURE in 1h 34m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1d3a158bc7fa412d8d6dff1ad5e522eb : FAILURE in 39m 01s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c096a1526f604e9790784a6173b12101 : RETRY_LIMIT in 49s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5130eee842a24ab6acd5a61c745daa84 : NODE_FAILURE Node request 200-0000020959 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a452ad87c73b45c9bd8f5bc8d10b6e51 : RETRY_LIMIT in 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9a79950405c447a2a7127beb33031c5a : RETRY_LIMIT in 45s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"5ffbc21b362f1f45bdebc2dc7724a868331a91bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 08:40:46.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e151bf6f068c458c8a0204e5c7c2ed80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f209b3dec6114d188108b9af7c174bcf : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3396982a987f47c08145079a83273e76 : SUCCESS in 8m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9e73014a8164aec9b234d32812dbd1e : SUCCESS in 14m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6a5f8819749465f8ff6d71a4fa74009 : SUCCESS in 7m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/74a2fb7b08504e5f85347dcbbb3c6edc : SUCCESS in 39m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6f2d7530939489a91a6e3c4811a8568 : SUCCESS in 35m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c782fc1a025845c49ccc786587acff0d : SUCCESS in 1h 38m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6149c2663e547d08f07eb55609b77eb : SUCCESS in 40m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d62f2827bfe649a9b07c7e371f115955 : SUCCESS in 15m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1ee85414d6f6453aac296ec6cf7d8c4f : SUCCESS in 22m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/77a6cf8b1d1749ba9b3fc749eae3298c : SUCCESS in 6m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4964e00ce30747b2b3e09233978c1a9c : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/66e425f5162c42e6920240130e0f19e8 : SUCCESS in 4m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/341cdde8df3f4abcab288966009967b2 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/afc9788cc86d4e63894c322b460ff6eb : FAILURE in 25m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0de36db32e364bb0bd6789a546f9be5e : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee97393e5f5d474b8d762458f14b93ba : SUCCESS in 2h 10m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f33f9e35c20a490191c6f8cc6fd06674 : SUCCESS in 2h 27m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6973d1be5bff4fc9af9f5896996d3f31 : SUCCESS in 2h 04m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/796e4b0c49a14a8dbdd9fd16927ac3f8 : SUCCESS in 1h 58m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fbb4b13acac8485685ac87a482b8d2f0 : SUCCESS in 30m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/762c9f58fb7e4d99ad37f48ea311e402 : SUCCESS in 1h 13m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fda6bee13656471392c83636694510db : SUCCESS in 1h 36m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/19a3a50f8dd4494184413b59137f466d : SUCCESS in 57m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c184dae281f481e96aa13f8d19dea04 : SUCCESS in 1h 22m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a8df9eb01c0a4395bb80b1426e6826f8 : SUCCESS in 1h 38m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/21bb4b87a9f9428d9cd292ead1413268 : SUCCESS in 1h 44m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38474a0038934eae98ecd74b303a904f : SUCCESS in 1h 06m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b9a55b5d99f43a99e16e96970e299c0 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":12},{"id":"7c37e2f591a5ca53a555bd57996152cebc7be774","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-07 10:00:08.000000000","message":"Patch Set 11:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/11/fc : SUCCESS in 1h 00m 57s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/11/iscsi : SUCCESS in 1h 19m 04s","accounts_in_message":[],"_revision_number":11},{"id":"6ee5a62fe86e811dca42d329fc425a1c20f970d0","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-07 10:00:11.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/12/fc : SUCCESS in 1h 04m 36s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/912051/12/iscsi : SUCCESS in 1h 17m 56s","accounts_in_message":[],"_revision_number":12},{"id":"a59875de89574b2c7af9e1a1b11e85fd8ede293f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-07 11:05:58.000000000","message":"Patch Set 12:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/12/check/fujitsu-eternus-dx-fc/f243fb2 : FAILURE in 2h 40m 05s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/12/check/fujitsu-eternus-dx-iscsi/88416d8 : SUCCESS in 2h 31m 08s","accounts_in_message":[],"_revision_number":12},{"id":"22f2085459bd4008fa038524663cbe2a66788225","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-07 11:29:38.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/51/912051/12/ : POST_FAILURE in 1h 10m 44s","accounts_in_message":[],"_revision_number":12},{"id":"f3cd2efddcdedce8df641227b1961c3a8e444339","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2025-01-07 14:29:27.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/912051/12/check/DellEMC_PowerStore_iSCSI/8ddff90/DellEMC_PowerStore_iSCSI/1833 : FAILURE in 1h 19m 25s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/912051/12/check/DellEMC_PowerStore_FC/0c2251f/DellEMC_PowerStore_FC/2029 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/912051/12/check/DellEMC_PowerStore_NFS/3b6de25/DellEMC_PowerStore_NFS/1525 : SUCCESS in 1h 19m 03s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/912051/12/check/DellEMC_PowerStore_NVMeTCP/7c2db4d/DellEMC_PowerStore_NVMeTCP/None : RETRY_LIMIT\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/51/912051/12/check/DellEMC_PowerStore_NVMeFC/edbba3a/DellEMC_PowerStore_NVMeFC/349 : FAILURE in 39m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"ec0736ea61a841e602a7ee2b909d8de089ade304","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-10 01:36:55.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/12/20250110062729/NEC-V-FC : SUCCESS in 3h 57m 05s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/12/20250110062729/NEC-V-ISCSI : SUCCESS in 4h 09m 23s","accounts_in_message":[],"_revision_number":12},{"id":"630e56015f2a51d8c67c28c2b012b8b608f83616","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-10 16:23:13.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/12/20250110193741/NEC-ISCSI : SUCCESS in 5h 20m 47s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/12/20250110193741/NEC-FC : SUCCESS in 5h 45m 29s","accounts_in_message":[],"_revision_number":12},{"id":"53fd67e714123649bd28d73b49739258cc37a9d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-14 02:50:45.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_01_2025_02_45_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"0d6d722e1c3b9e73354ed7c566bf775dc4064b7a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-27 13:45:47.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"95a4b36aa68999f092b16c3fb00f12c7395599da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-21 07:16:24.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"9d0e03a629ec2382f65ca71e9c9459cd0c903bfa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:16:33.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/13/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"49949559dff120bc028f7d2ddc719ad35ca17b75","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-21 07:18:04.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"9cd922aba898adff77987d0d54db59e59aa8cc85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 07:37:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0415ad0318c54d8f947c651cd9409dbc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e83034f22f364afe87baa29d60ec1049 : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a249b3b0c475474395b48409d80e8740 : SUCCESS in 18m 20s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4c8a073ea4b2a7d175447c1d128e41315a308123","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 08:42:24.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/481/.","accounts_in_message":[],"_revision_number":13},{"id":"eb0aa638b437d12d9a0c113d52cb9da5f3cfbe48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 09:49:15.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba86c02c5b7f4aceab70c4c6c1460ecd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea88850d38f43aa95762b8aa2472507 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7824f6e2848847909c519ab4bd005127 : SUCCESS in 10m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f82972712ebe4a25a38e88e6b3ae1161 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6335ed4e904244dbb6fdf32a6c5cd2a2 : SUCCESS in 15m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/10abc75c8e5c46e6a1300fb04eaaf269 : SUCCESS in 59m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7c64901f0094567a8d0b856a93a55a3 : SUCCESS in 50m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3df7207beab9408ba8b685bfe4bfa401 : SUCCESS in 1h 57m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c79ee1e5ec524b8192f1d3fdc3f115f4 : SUCCESS in 1h 05m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd74a202d5ce4d96b3c52d5a830af90f : SUCCESS in 22m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0e812ca662a44eb3a4074df1625c62c7 : SUCCESS in 32m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8bd0e0a9927f4509972d5996c9e51e78 : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/18b14ea16fe44aba8fa8637818e11e5b : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e55a8138f2c944ed94bb73a4efc407d9 : SUCCESS in 8m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/38901bd9e9c14fe3bc659f48e594cae2 : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c23512f97b43464c823a31281c051d44 : FAILURE in 17m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8788ae23d9c143338a1019bbbdc45524 : SUCCESS in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ea5a05bb2de4ef9bd66d78b032571e2 : SUCCESS in 1h 00m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b8019b9985674f1182818a8651832d19 : SUCCESS in 2h 27m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aad0315fec8a44e588eaca9573d86ce7 : SUCCESS in 1h 54m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/99c1fdad4b204eacb981470dc5adc696 : SUCCESS in 1h 42m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d57b56884a43409b98467a452223efa5 : SUCCESS in 30m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/866e344a7b7c41ddaaf227932efec1da : SUCCESS in 1h 16m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f84cc63c7c314110ac8fe2cb975ee60a : SUCCESS in 1h 34m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dbffdc285a0347728179373e300f4e6f : SUCCESS in 37m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/117fbd565d6d4568a294ab7134741ed4 : SUCCESS in 1h 46m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c2d3cef1761c4f3dbae43c2aecc3aa2b : SUCCESS in 1h 19m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cadfafeae1cd4196b7baad688df90d38 : SUCCESS in 1h 38m 28s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8399fe4b3e8b4db3b01f1333a0c8034f : SUCCESS in 1h 45m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e02fbf700aa342ba817306cf30b88c61 : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":13},{"id":"63351d900f0acad3ac7e543ba74b2b6c79155c08","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-21 10:37:38.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c1d5475750cb49c3ab457c9f0ab3ff9f\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43583cbc0a3241ccaf3c469b125f20cf : SUCCESS in 45m 10s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dc652870d3194f169af6041dec0a62d2 : NODE_FAILURE Node request 200-0000033674 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/168034fade5b4db1a353506e0053e52e : FAILURE in 2h 05m 50s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5633c9484d324e248241aebb27a800e2 : SUCCESS in 43m 43s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/05b83ec48b36407e9cf6a15dc7dbb8fb : RETRY_LIMIT in 37s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/937e43d1d9654fb1b3ff3fdf28c6be64 : NODE_FAILURE Node request 200-0000033678 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ea9e3911d3a4b459e9b4a624e24a762 : RETRY_LIMIT in 38s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4332874ed31d4fa4824d0d52659ced6f : RETRY_LIMIT in 36s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n           ... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"4436c0e284ef751f9cce5c0d4952c5faa9ca9c19","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-21 12:13:08.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 34m 10s","accounts_in_message":[],"_revision_number":13},{"id":"5105ee88f90c7df75b42bcee8e2dc1b86f5bf47a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-21 12:19:31.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 35m 00s","accounts_in_message":[],"_revision_number":13},{"id":"35af1ea3d269dfe8a3a8e0ce7273c63ab365ee0f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-21 13:10:47.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"b0dc8263bc3ae133884d7e9f2a4e65013a44190e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 15:13:11.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/550/.","accounts_in_message":[],"_revision_number":13},{"id":"9bdef849db76c8527f59c287118a40167c4ba619","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-21 17:59:42.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"b571dd27ee05ae16362c496aee5057e908684909","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-21 19:27:28.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9057a7d653564ad5a8b633d50a61c5d3 : SUCCESS in 1h 12m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b3cb633e7efa4a138af08596a1376ab1 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":13},{"id":"f1a48944bfddb6fdf714e8ecad53a78bfc6796f7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-03-22 09:12:53.000000000","message":"Patch Set 13:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/912051/13/check/fujitsu-eternus-dx-fc/f03c348 : FAILURE in 2h 27m 53s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/912051/13/check/fujitsu-eternus-dx-iscsi/f7cfa52 : TIMED_OUT in 3h 38m 14s","accounts_in_message":[],"_revision_number":13},{"id":"5ffe50388344595331628be670d32e9eff917079","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-24 11:21:52.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/13/20250324160540/NEC-V-FC : SUCCESS in 4h 15m 24s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/13/20250324160540/NEC-V-ISCSI : SUCCESS in 4h 16m 09s","accounts_in_message":[],"_revision_number":13},{"id":"ae395d608fd735efb9f8075df407573cb5442ca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-26 03:03:00.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"56418f8c7048dfb46e7cbd0d5fd9db06465a468f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 03:03:09.000000000","message":"Patch Set 14:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1575/","accounts_in_message":[],"_revision_number":14},{"id":"16c2a048a90b660d17d68fa9dc122a429f4a4fa4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 03:03:12.000000000","message":"Patch Set 14:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/14/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912051/14/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":14},{"id":"1d9e0e916bc824dab3e914b6c7d3a2fc0e1abc4c","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 03:04:45.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"d2f20837fca952200336a930e5ed1a5c02e50c08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 03:38:40.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b893f8bc4845407fb8ad1e944edc7746\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/362d786997b64bc18b36ed6c0078096f : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e71501efc8ac4681963b48a6c7a08e51 : SUCCESS in 15m 06s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8ade040f80e79ce16dee3a9f3586358eb4294f7e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-26 05:31:30.000000000","message":"Patch Set 14:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-912051-14/FC/ : FAILURE in 2h 28m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-912051-14/iSCSI/ : UNSTABLE in 1h 7m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":14},{"id":"46f9c865a0663bc207ef3ebf50e1c7dd19b694ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 05:34:17.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/748e980cc8ed4a66b90385896fc89917\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4d16e42b7049afa2d618df01fc5f1a : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0cc4479b30240d8a2158667f08998d5 : SUCCESS in 12m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/169ad393b7b0476faa2c64bfd2584c4d : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711beb1aaafe4a4c9c595f483c9bf22e : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/3aee1665cdfb4071a393d46429ee361e : SUCCESS in 47m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/077d08b27f1143cfb27963621280d4bd : SUCCESS in 1h 02m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03193de4e895408e85d214813d1fbd7f : SUCCESS in 1h 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2701ec51d4e473495651b6835d27053 : SUCCESS in 1h 03m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b07951389c6424b9d82198e8703c725 : SUCCESS in 23m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7e1a8447528e485381e8d52135ba382e : SUCCESS in 31m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1b10f6f738c34006ae6ea4ce72a77dec : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36784e12e8db4c03b6912b018af76c17 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9d4646ecba024703a848c1e24116c435 : SUCCESS in 7m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6ef780eeee2a4876afafcf0d0ece3b40 : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8342a5e370014996bb311951b39c9acf : FAILURE in 37m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/365e288685bf4ba49465ca5e9db46d2c : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2784a69e01064819af6ed1df16b29a0a : SUCCESS in 1h 22m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0038e857eafe461091aafa2bcbb48da1 : SUCCESS in 1h 48m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4f71df85ded9456fa336b43b0e5f5eb1 : SUCCESS in 1h 48m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5abcd7fb07f04227a88b32214fa03768 : SUCCESS in 1h 49m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a5f8d6519dc344ca9081dc282cce5d8d : SUCCESS in 34m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8cd31b5feabf49d4b75d6ca0c633a107 : SUCCESS in 1h 05m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/72b5a03d3c3f41ea8e81683b701047cb : SUCCESS in 1h 32m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/136faf766c214f2eb01ad92c463079cf : SUCCESS in 35m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fda68cf995f94f70ba9278c7754a0db2 : SUCCESS in 1h 27m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/480631d293cb49bbb6d245c90fe28cf2 : SUCCESS in 1h 35m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/80da123d57964616be75057bba4d2990 : SUCCESS in 1h 35m 15s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c2570e632734ac6a8400fb4fc64b6b3 : SUCCESS in 1h 14m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec72b8be21ff4982955ff607e9d1772c : FAILURE in 34m 58s","accounts_in_message":[],"_revision_number":14},{"id":"c860ea2d614021d4eb40e71f52c4cee15ed10eb5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-26 07:27:07.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/23e83e2a07f543e9ab82bf7df93fbe79\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e8bb7dee6484e4bb5f2d987df8f6ed3 : SUCCESS in 41m 47s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/019099b05db14e598f1c5bf6367ab70b : SUCCESS in 44m 46s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/810268c7000b4a589fc098533fcc2dee : FAILURE in 1h 26m 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/36d53d1450b3459ab2beff06665bd891 : FAILURE in 45m 05s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/526c5db0b1394842be590a5f62095d53 : RETRY_LIMIT in 36s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e002f76c193c4d609d41e43e73b3d49f : NODE_FAILURE Node request 200-0000035399 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3aedccf127b647d7bf5cec004f25214d : RETRY_LIMIT in 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/207f40a4805e4bde81212d4a1e602b07 : RETRY_LIMIT in 38s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"d1f66b16a860fd17608239bc9c6049f1afff9ce2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 09:47:47.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912051 : FAILURE in 32m 51s","accounts_in_message":[],"_revision_number":14},{"id":"99f9559e198b4c59ef26f257c1df75d9a4e06cc9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 10:06:12.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912051 : FAILURE in 33m 45s","accounts_in_message":[],"_revision_number":14},{"id":"81f0bb1df2e9bee3992a4d6ad9c9ecdc72b99958","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-26 12:45:14.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"012216b34eedb67110c7b57437e80e92e78a679c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-26 16:26:40.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/578/.","accounts_in_message":[],"_revision_number":14},{"id":"e9e437c2e4740939b6e6f03ff6812551de561860","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-26 17:07:49.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7b88f4da6b96496c942d8d9a85006959 : SUCCESS in 1h 17m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7640ef2c09b345e7b10d1dfd413f551d : SUCCESS in 1h 21m 57s","accounts_in_message":[],"_revision_number":14},{"id":"dc5e6b0b785a139b137fd96d351086bfa8a3e0a1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-30 19:37:37.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/14/20250331003812/NEC-V-ISCSI : SUCCESS in 3h 56m 05s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/14/20250331003812/NEC-V-FC : SUCCESS in 3h 59m 23s","accounts_in_message":[],"_revision_number":14},{"id":"d34b57b325b74714233c1569aeba6519e03f1ddf","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-31 18:35:55.000000000","message":"Patch Set 14:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/14/20250331153047/NEC-ISCSI : FAILURE in 11h 45m 56s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/912051/14/20250331153047/NEC-FC : FAILURE in 12h 05m 04s","accounts_in_message":[],"_revision_number":14},{"id":"22a8906c2489a6c62f67cea6907a4a88dec53a3a","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-04-02 06:06:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e57f763ad0923b5eabcbe6626db5df208ef73f83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 06:23:17.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba6b386329114ff99ce8953e047bc2f6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1bc3cb07b5d1410dbadeb24e16c91735 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4d485140952473cb5d302275ea9cdec : SUCCESS in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"201dabba9872e3c77ced564f85427f48663293fb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-02 07:26:39.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/890b442eb4f847e585de9fd7e1ca62a2 : SUCCESS in 1h 07m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e1c57dd7a109484eba05e70cf9f3c5f1 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":14},{"id":"3abd6e495e429f4350c9827518053ea52ab48381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 08:44:02.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d481ba5fd11046cd9ea85f25486058b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa00d1b8f4e0440198f892160ea5cba6 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6fae14d15c6442582f3485bfa2b4134 : FAILURE in 10m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6896c276367f4997b54211bd3fec1124 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/008f3ef580fd458aa2ddf35bc2cfaecc : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f49aee78eb6648b7a9aa1ee84b97cd6e : SUCCESS in 52m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99b1688b8a64418ea16d28839009e340 : SUCCESS in 54m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f53b80d50daa455386d844e5427d7777 : SUCCESS in 1h 41m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d7e3032c3ab44549721053315aac6a0 : SUCCESS in 1h 02m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdc210f1596a4b11b0e5373ded241947 : SUCCESS in 21m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/721109e826c44d66ab980d96937e3fd5 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b0ead6dc1add4a18b77fb4a9d74186f4 : SUCCESS in 6m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de67abd3ff4a46a09dc0d0ee4d80cb09 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b1810313b9a84e09a5873fc96a033bbe : SUCCESS in 7m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/accd02cd26c84ff88c91e54394c6304a : SUCCESS in 4m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2ee55cfaa6b44a9587f9e742837f5c51 : FAILURE in 38m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8472b67e4f5041bca048298551c35dcc : SUCCESS in 6m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf8c0f9e0f4748f0b42b47e965beb921 : SUCCESS in 58m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/81e17a69e76a45cea55925ca1266189d : FAILURE in 2h 24m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/153ab228b29645fbb51f59d7201becb0 : SUCCESS in 2h 03m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d17c21e823d747fe8de5924e7c3f7b47 : SUCCESS in 1h 30m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fc551802c407473894ed20540ccc348b : SUCCESS in 24m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fdaeb24e7cb24c638f50bd9fbf032212 : SUCCESS in 49m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/212daf9a66aa426c989a9a59b8384f5e : SUCCESS in 1h 20m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c454d10f96c9497c8fd4b57760ae018b : SUCCESS in 23m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e78f81b054343db954fc0f6a38f0a9c : SUCCESS in 1h 44m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8abae84a7c924abd957a7bbd7862ae6b : SUCCESS in 1h 09m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bece52060aa4ccfb275017a14e94af7 : SUCCESS in 1h 17m 10s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55dafa54ba5648edbcd544a42ac2db27 : SUCCESS in 1h 25m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78a89a25d4624b608b02028b779b3073 : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":14},{"id":"8a91e469a9af74171cd1ad7156c9e939c548305c","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-04-03 00:31:19.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a53cf8a8fe26271bce581051018aa35dba130274","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 00:53:42.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cebdd2119ee4140890af01a5d381e2f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30073d70303f43ba83baeaeb3daf3580 : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b608577c2ce4144941f3212fa51955b : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2da9a1a6d8014a33b0ab7651bed74c42 : FAILURE in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f8275776b514e7c56cd77475fc48e2ed4ebd82d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-03 01:56:51.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0a3c148e8dc345bd893845a8e53b3b15 : SUCCESS in 1h 05m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/708521ed0107441190211bf682666967 : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":14},{"id":"d53098226cc133ed6600497325de0a3ebf1b1780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 04:07:26.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c03f9ddd730d4ecfaeb55ed8e7a05a14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/166d6adb0d9a41659b2dca9ff3970cfa : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/650b445930044ce3b6bc6d0ea4cc4558 : SUCCESS in 9m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da7f1753b06548199db07f15ebd14653 : SUCCESS in 16m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8f36fcbe53d430fb0e718bdbbebbe5f : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262993b3c1ef4351b30116434d6355f2 : SUCCESS in 15m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/d6a572ad9c364e51b6513a3dffd7831f : SUCCESS in 1h 03m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d20da1ad76f644ce9e56d7eb373edacb : SUCCESS in 1h 00m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f208f9d7ab041209f813ced7e6ca8b5 : SUCCESS in 1h 45m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f54b2ed6b34748a9a59eedb6f660496e : SUCCESS in 53m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e507ed8d48a4890a5d2c50b35f243f9 : SUCCESS in 19m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fde852ca9fac43d98159e3d82c418ad3 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/099e77a14182460bb3016f215bcb202c : SUCCESS in 5m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1fffceec6e5470d9735c0479598c547 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c15b3bc0fbe745dc9f36d7b7c82d767c : SUCCESS in 7m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/538b325ecef7423793555b6dcc07cd41 : SUCCESS in 5m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f5338f88a2cc4f59a3678f0a81518a50 : FAILURE in 34m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3b7d52e2f134805aed3965b18f2e8d0 : SUCCESS in 4m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1cfdc3f5ea394e5b88c941a09c2da09b : SUCCESS in 1h 01m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/08e175669c3048c1bf6e33d0713b7997 : FAILURE in 1h 21m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a929d774b06409fb1e4033e97b48ca9 : SUCCESS in 1h 30m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/640d08195b1946aaa182ce03679cb91c : SUCCESS in 1h 57m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/be98d709c98c47cd8ea953239edf05aa : SUCCESS in 22m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b54323d8a41f49568551a5b7d84eaee6 : SUCCESS in 33m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8323d9c46780442b96e02755dc7a7359 : SUCCESS in 1h 37m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2902a6aaef89403da638bb59e191efb0 : SUCCESS in 27m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bc82ae1c60624b9386ca2089254d2a43 : TIMED_OUT in 3h 34m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/88efdbb96a874127ae897458c9a9b39c : SUCCESS in 1h 09m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/473b66f07fea4a5f9f90f882f5aacdae : SUCCESS in 1h 24m 56s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5564ea01be64aefb4477b632bb9eaaf : SUCCESS in 1h 08m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/325bdd99a331425c9bd29186f1dc1e90 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":14},{"id":"f8cb540f95d1be33a6d88d32008c28b8664c37be","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-04-03 06:21:07.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1b721925ada5faaa3322d78e5100740fc2ecd588","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 06:41:24.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b8a8aab8ee4337b74b183ff0ab57be\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/471333d4211840168b42605a3465fc29 : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd5332cdb733412d8411a42dfbadd3a7 : SUCCESS in 13m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3da4244a12a042909eeb8929b804ae5a : SUCCESS in 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"483ef1897022fac8abe81b0864a7f27a28970c3d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-03 07:40:58.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/29fc5e33a9e14b43b745f8f866528dd7 : SUCCESS in 1h 15m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e65e13283fa14623b5788634e4ebe092 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":14},{"id":"551b26222649a8f317b9a1c40c72a8a360484937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 08:53:53.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d19a0e24ae904ebcba1c0576dfc886bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5fd610038824f11a3b799d8a2f6b2ee : SUCCESS in 8m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/218aec20ab8a4527a55d210fb8d0b0d7 : SUCCESS in 9m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e924537eff4e4d31a628cd6fe4ae5d8f : SUCCESS in 14m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b7042a9c4bd4db183342b4743721ca2 : SUCCESS in 16m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2cd0037dad5472aa0fda171216a72ef : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/0fbc1396e37a4337a6418b9e84603600 : SUCCESS in 56m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ac9c796b61e49d1bc5ef5b26d0d5389 : SUCCESS in 27m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e89c356d0320480299c2fd6eef8b67dc : SUCCESS in 1h 27m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a1135c06a5b4e2fb7201f1ea09ed1d3 : SUCCESS in 31m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f914f88c795486994fec70273938f1d : SUCCESS in 20m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/395a58b7e4d047abb2ba1bbd5a18f33c : SUCCESS in 26m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c459a807869d47da9c6dc67685978c97 : SUCCESS in 3m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eaec9933f676473faba8dc7a1a0a1d94 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b70d85db5c7745a6ab01272b7cf68b4b : SUCCESS in 4m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/40e4acd4122e4800b4a0ecc021d0a242 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/48154f0353144bdd85935f200f3e5900 : FAILURE in 43m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8d437fe2dc4b45098e1f096b7b5f0c68 : SUCCESS in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4aa14f2cccae478192edcaebe56fd77d : SUCCESS in 2h 07m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/42b3e76555cf4a8794035e093b7a3ac3 : FAILURE in 2h 24m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2ae28ce3af0f41a0af0eb123a122ff13 : SUCCESS in 1h 39m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b1af56e3330d44d0b19504807fb0c3cc : SUCCESS in 1h 32m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/70951258c2004fd895c212fb7b206688 : SUCCESS in 17m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5998cee423cc4768be4db40f6204c9e8 : SUCCESS in 1h 05m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7751ae5314b44914a093aef17a2e5fec : SUCCESS in 1h 47m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9e02893af52745bd91ce3b8e3409455a : SUCCESS in 37m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3bfeea0412fb4beeb5619da57fcab906 : POST_FAILURE in 1h 52m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/33d24765b11b4e229cc46de76a1d16a0 : SUCCESS in 1h 45m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8c5d01d87f54c2e9bfb3037659598b8 : SUCCESS in 1h 41m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21137737c1554df6966797554fc7b8f4 : SUCCESS in 1h 26m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e9ff761d3164626a261d4a5bac201ea : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":14},{"id":"76399d018873ca11dcedcbd22963117bc097526b","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-04-04 00:24:13.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5497ff21c5f94ba38344a215c51ee9436ed3f4da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 00:41:47.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f360987e48344a1bc127c8d3d9473d4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8510dad64aad4d71b6073db1a3e6e0df : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/93c9db9d40c040ffbcc87422d2f124cc : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1647102b9ffd4509a7f51f3835c61a13 : SUCCESS in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e37afb3f2740b6c58430c785fc558715d3ab8f70","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-04 02:42:09.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2ef938a022914323bc375a7e9a1fe611 : SUCCESS in 1h 11m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/261d20925e2349239035e652e7043f64 : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":14},{"id":"83c3d11363f5007105f5e05ef1ba8c8bdf02d8f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 02:42:29.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0bc70a63a5b45bb89b423e73aacec54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47bd458e3cd54aa1b4cafb289bb57cde : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f5163e34f1848ae906d4da72a515806 : SUCCESS in 9m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18e67e1597084915a5c21bfc162ad91e : SUCCESS in 10m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/424b1d71c11a484999572f3d6e65a096 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a123f825824f45b9b205f2a04d61ef13 : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/b0d917a5236248238ed193d2ed7ac0a3 : SUCCESS in 45m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02b405615a394567b7d0aa7f4d8b3493 : SUCCESS in 42m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a3775b363994ff4b7f71245ef71138e : SUCCESS in 1h 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c1cd0e0e8454b149666f5768f46f14b : SUCCESS in 52m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1571b60ced1470b906bfa00086d048e : SUCCESS in 15m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f2bf35f280f943709be67090c1e4d3de : SUCCESS in 21m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1596bcffe3b64906b82e5a923663dfc0 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/baf4a39b0bcf40b49df5089557444734 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/074aa560b6e84de7ad7008a7c21ff09e : SUCCESS in 5m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7e0cc7ad7a5f4376aa87c4bcfff2ba7a : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dd711058961949bf9e338a6f9ffba40c : FAILURE in 26m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/107c69bed2334ca4ba39134ca8d9203c : SUCCESS in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2c0310ce100446efa3415b18ffdc366b : SUCCESS in 1h 59m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1cea882544a34e8c8a29a03884db19c6 : FAILURE in 2h 10m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2c2aedf063a04ef08003f97bf542f6b8 : SUCCESS in 2h 03m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2921e1e87cbf4a17876c4868b8ae3955 : SUCCESS in 2h 12m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bd81054cb6104c7f91040553c5c2d758 : SUCCESS in 24m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cb1954bc50044194830056de24109251 : SUCCESS in 1h 08m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8928bd607a2b4eec84733c3f236052fc : SUCCESS in 1h 52m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7010e67f40bf4bcc80802b201972c015 : SUCCESS in 42m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/87a09e0a1b5e4a5bb2305362b58d3501 : SUCCESS in 1h 07m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8728b4f921a948a19f3296d24e65d618 : SUCCESS in 1h 40m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ea48ab7f680460ab84ae22b15d79c9d : SUCCESS in 1h 21m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0abf3dcf89bc43379bea52a22ae30992 : FAILURE in 1h 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7c633f21c1401385bc0852fa077a4c : SUCCESS in 47m 31s","accounts_in_message":[],"_revision_number":14},{"id":"c019f991de841690e6e3f340b6d07b26abbbcb83","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-04-04 02:46:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"14789c327184ee89c33c8e0bf72674e75f8d0d53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 03:08:28.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be2ade741ec64376adc8386b2c8ce669\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8c8c348cb824c6ab8889a5716b13c07 : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e4445c5537647059b1d87a1ea684266 : SUCCESS in 13m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/204912d93e304e2591ce5017cee92924 : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"60a9af46e7bdd1f4ca4387d530b19a30e209232c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-04 04:06:42.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2deded8591d045a1874603d91e677b41 : SUCCESS in 1h 09m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cf33776e04c34a6795bf3ee291f1a9fb : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":14},{"id":"09892cdda076c59e90d085a3348b169b27480766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 05:06:27.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f37707041f34cd6a2761a173bd511a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666daf9389dd482e8b216c9f5b259caa : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b5b44f5ad3d4549bd820ea2351d69da : SUCCESS in 13m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f152ebbeb894efba9ca21f244df853c : SUCCESS in 12m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f43e43293dc4686be9b1ff7bd68e745 : FAILURE in 15m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87a1815a49e4e70a1f47dc40349afdd : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/8f2c4fce5992440fab99acfd44c0427b : SUCCESS in 55m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b31ad025229441981f8f788eb50fa4c : SUCCESS in 56m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ce82016f85746d89a09f8eb53ffe5b0 : SUCCESS in 1h 35m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/235816eea0b540ada673ace625c6d7a3 : SUCCESS in 53m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a977262453143bb83edf35c7dcf48b7 : SUCCESS in 16m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8dff684760254ecbb049ebdaba48f8a2 : SUCCESS in 24m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5a6d347456504787a63ad92dde1c8f97 : SUCCESS in 6m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79f9a445c52948758d49c9134c32aba6 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7d1ca4cab3e34aa79b21f260d8a0e348 : SUCCESS in 5m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3e11de84f80d4961affcb2bdc8f0a516 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/122f5f4aa14a411d9b0b99d8f3974e19 : FAILURE in 28m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52c380770dbd4247912e97b2ae2b99a5 : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/489a014070a44f2b8e0e25971bf6a435 : SUCCESS in 1h 51m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3646cafaa4404d39a86c391ea49f8cdd : FAILURE in 2h 17m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a8049db2284343b386d5fb146a9937f2 : SUCCESS in 1h 45m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7a3772e47eaf40ff897434a77b9df523 : SUCCESS in 1h 44m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a8fc1ea1db79448cbaf79fbf593866b4 : SUCCESS in 30m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7f41cc5bfb764da68b1e45f7e1d029e7 : SUCCESS in 58m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/62ef2c75b3a6491b8e68a4215247d8b5 : SUCCESS in 1h 21m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1c7db45de932485688f9aff3d71f984d : SUCCESS in 38m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6eb66f889ef5423791bcba5eb14826cd : SUCCESS in 1h 38m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2347c805a32e4d02b2c2596d3de40aea : SUCCESS in 1h 30m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45e1036f702e4510b6300236ae640ed3 : SUCCESS in 1h 39m 27s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2669ccad119f4664b4cf19545305699e : SUCCESS in 1h 24m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e992f4551aa4d51a527152efad59fd6 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":14},{"id":"3290ff2dfaa4263080c8bbb8865628497ad2d607","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-04 08:14:45.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"49a71800728b9f4db7c994e07209f1b33ef884e9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-04-10 02:35:07.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/912051-14/PVME-iSCSI/ : SUCCESS in 1:57:02\n- PVME-FC https://os-logs.seagate.com/dell-ci/912051-14/PVME-FC/ : SUCCESS in 2:06:42","accounts_in_message":[],"_revision_number":14},{"id":"8253803c0369cc5060e8f1da56d040c28ed88876","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-10 07:02:16.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/912051-14/Lenovo-iSCSI/ : SUCCESS in 1:59:19\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/912051-14/Lenovo-FC/ : SUCCESS in 2:06:53","accounts_in_message":[],"_revision_number":14},{"id":"f9c3a24050c9dc80e55a6b96271547061879fe02","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-04-10 11:28:16.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/912051-14/Seagate-iSCSI/ : SUCCESS in 1:58:18\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/912051-14/Seagate-FC/ : SUCCESS in 2:06:52","accounts_in_message":[],"_revision_number":14},{"id":"0733d76c0e4a4e07eef45d7abe1ede801345ee70","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-15 13:52:40.000000000","message":"Patch Set 14:\n\n(6 comments)","accounts_in_message":[],"_revision_number":14},{"id":"489225a2a4b0e4b75b47be5c3f21ac45c6f14bc8","author":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"date":"2026-02-02 08:50:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"a6616aa9d45dc64a8cd89c718447b9db55573ed4","revisions":{"fa71fd44ed1b0a7dbfdc714dcb730f3bc123b693":{"kind":"REWORK","_number":1,"created":"2024-03-13 02:44:13.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/1"}}},"commit":{"parents":[{"commit":"cbb73db6fabda3704e8c24519359c7cc2c4cdf61","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cbb73db6fabda3704e8c24519359c7cc2c4cdf61"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa71fd44ed1b0a7dbfdc714dcb730f3bc123b693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa71fd44ed1b0a7dbfdc714dcb730f3bc123b693"}]},"branch":"refs/heads/master"},"6dbe67a0f0800bbec81d64cb7b7451324e30d0fe":{"kind":"NO_CHANGE","_number":2,"created":"2024-03-19 06:26:06.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/2"}}},"commit":{"parents":[{"commit":"8f025b8bfaa10a13f6df0df4d79e5490abd204ff","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f025b8bfaa10a13f6df0df4d79e5490abd204ff"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-19 06:26:51.000000000","tz":-240},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6dbe67a0f0800bbec81d64cb7b7451324e30d0fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6dbe67a0f0800bbec81d64cb7b7451324e30d0fe"}]},"branch":"refs/heads/master"},"97b02ba37771e93e6ec997b19d8d2d506f822b29":{"kind":"REWORK","_number":3,"created":"2024-04-08 09:34:16.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/3"}}},"commit":{"parents":[{"commit":"9b719e56710966dc9f5cf2d89aabd16e099da589","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b719e56710966dc9f5cf2d89aabd16e099da589"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-04-08 09:33:35.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97b02ba37771e93e6ec997b19d8d2d506f822b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97b02ba37771e93e6ec997b19d8d2d506f822b29"}]},"branch":"refs/heads/master"},"deb3b039893661d061ce88b7d7a74e161c594f27":{"kind":"REWORK","_number":4,"created":"2024-05-21 03:54:38.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/4"}}},"commit":{"parents":[{"commit":"7a76cf77a8e6706ede84940e4e3056d1a48477be","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7a76cf77a8e6706ede84940e4e3056d1a48477be"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-21 03:54:08.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/deb3b039893661d061ce88b7d7a74e161c594f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/deb3b039893661d061ce88b7d7a74e161c594f27"}]},"branch":"refs/heads/master"},"a43d5e8850a0f1443ee2d3c05a464c032fb19fd8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-24 10:05:02.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/5"}}},"commit":{"parents":[{"commit":"b4c36a00186cae88ad48ed7808b724430208b7e3","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4c36a00186cae88ad48ed7808b724430208b7e3"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-24 10:02:23.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a43d5e8850a0f1443ee2d3c05a464c032fb19fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a43d5e8850a0f1443ee2d3c05a464c032fb19fd8"}]},"branch":"refs/heads/master"},"5063a7faa869c4d7214bd7d6f8cb5d140c3484d4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-27 09:50:14.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/6"}}},"commit":{"parents":[{"commit":"13169081428fc9ce9d6e87187130549b7407e35a","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13169081428fc9ce9d6e87187130549b7407e35a"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-27 09:49:39.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5063a7faa869c4d7214bd7d6f8cb5d140c3484d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5063a7faa869c4d7214bd7d6f8cb5d140c3484d4"}]},"branch":"refs/heads/master"},"cd50354dd7f2edca10dac9ae20bce7cd1f589b4c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-06-14 06:58:37.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/7"}}},"commit":{"parents":[{"commit":"3bc08c4fd00f97996711b90d1adf5ed66563b9eb","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bc08c4fd00f97996711b90d1adf5ed66563b9eb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-06-14 06:57:29.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd50354dd7f2edca10dac9ae20bce7cd1f589b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd50354dd7f2edca10dac9ae20bce7cd1f589b4c"}]},"branch":"refs/heads/master"},"721daca79c940ca9858af13f5a1b13417ebb628b":{"kind":"REWORK","_number":8,"created":"2024-08-05 07:31:39.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/8"}}},"commit":{"parents":[{"commit":"d1f38adce8b1575c9819927ac8967e88e7f6635f","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1f38adce8b1575c9819927ac8967e88e7f6635f"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-08-05 07:31:12.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/721daca79c940ca9858af13f5a1b13417ebb628b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/721daca79c940ca9858af13f5a1b13417ebb628b"}]},"branch":"refs/heads/master"},"ab7501eeda68ca314f4b38132dc2f7472970f102":{"kind":"REWORK","_number":9,"created":"2024-08-05 07:38:38.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/9"}}},"commit":{"parents":[{"commit":"bcda8ed698b166a40d54b5a2f93854d5d8a51dc3","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bcda8ed698b166a40d54b5a2f93854d5d8a51dc3"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-08-05 07:38:15.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab7501eeda68ca314f4b38132dc2f7472970f102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab7501eeda68ca314f4b38132dc2f7472970f102"}]},"branch":"refs/heads/master"},"64551c802ee461bbe2c1b5525d8945d3f604ddbb":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-09-02 01:43:54.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/10"}}},"commit":{"parents":[{"commit":"831930f72ae572c66241bd26b832dc50d04f44ae","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/831930f72ae572c66241bd26b832dc50d04f44ae"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-09-02 01:43:15.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64551c802ee461bbe2c1b5525d8945d3f604ddbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64551c802ee461bbe2c1b5525d8945d3f604ddbb"}]},"branch":"refs/heads/master"},"6c28766923d1b41802609f233d171a33bda779d0":{"kind":"REWORK","_number":11,"created":"2024-12-13 07:18:32.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/11"}}},"commit":{"parents":[{"commit":"42eacd39659f41acd3c1cbd62ad0d9112c81d063","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42eacd39659f41acd3c1cbd62ad0d9112c81d063"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-12-13 07:17:57.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c28766923d1b41802609f233d171a33bda779d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c28766923d1b41802609f233d171a33bda779d0"}]},"branch":"refs/heads/master"},"97d7fa29a14321fd1ab45135b4982f810a456360":{"kind":"REWORK","_number":12,"created":"2025-01-07 06:10:34.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/51/912051/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/12"}}},"commit":{"parents":[{"commit":"d111c074c368b643732e925c2828b403af6e718c","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d111c074c368b643732e925c2828b403af6e718c"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2025-01-07 06:09:30.000000000","tz":480},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97d7fa29a14321fd1ab45135b4982f810a456360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97d7fa29a14321fd1ab45135b4982f810a456360"}]},"branch":"refs/heads/master"},"4b159dc5f8e668e8ada8098e46d5f7066eca74a7":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-03-21 07:16:24.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/51/912051/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/13"}}},"commit":{"parents":[{"commit":"a8a50ac5b8571b6e4c5027af9ca1e0a339b52936","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8a50ac5b8571b6e4c5027af9ca1e0a339b52936"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-21 07:15:25.000000000","tz":-240},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b159dc5f8e668e8ada8098e46d5f7066eca74a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b159dc5f8e668e8ada8098e46d5f7066eca74a7"}]},"branch":"refs/heads/master"},"a6616aa9d45dc64a8cd89c718447b9db55573ed4":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-03-26 03:03:00.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/51/912051/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/912051/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/912051/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/912051/14"}}},"commit":{"parents":[{"commit":"8e0ee389993681cdeb81d25a8eb6e4aa50471937","subject":"Fujitsu Driver: Change the function of attach/detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e0ee389993681cdeb81d25a8eb6e4aa50471937"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-13 02:44:03.000000000","tz":-240},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-26 03:01:44.000000000","tz":-240},"subject":"Fujitsu Driver: Add mapping file support","message":"Fujitsu Driver: Add mapping file support\n\nFor a host, previously only one host-affinity is allowed to exist\non the ETERNUS at most.\n\nThis patch adds parameter ``fujitsu_port_host_mapping_file`` is added,\nwhich sets the mapping information that specifies the port of the\nETERNUS AF/DX to be connected.\n\nChange-Id: Ia8efa574b36d04132a8e384f6f39fb2eee164289\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6616aa9d45dc64a8cd89c718447b9db55573ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6616aa9d45dc64a8cd89c718447b9db55573ed4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
