)]}'
{"id":"openstack%2Fcinder~912058","triplet_id":"openstack%2Fcinder~master~Ic22add411b594d90d12b94e85055a33db40c197d","project":"openstack/cinder","branch":"master","topic":"bp/dell-powerflex-nvmeof","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-05-13 13:45:01.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"31016":{"account":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-05-13 20:53:43.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"37328":{"account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"32912":{"account":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"last_update":"2024-08-19 11:47:55.000000000","reason":"removed on reply"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"36952":{"account":{"_account_id":36952,"name":"Karam Jameel Moore","display_name":"KaramJameelMoore","email":"wawmart@me.com","username":"KaramJameelMoore"},"last_update":"2024-04-11 14:06:59.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-05-13 22:53:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic22add411b594d90d12b94e85055a33db40c197d","subject":"Dell PowerFlex: add NVMe-TCP support","status":"MERGED","created":"2024-03-14 04:54:31.000000000","updated":"2026-06-19 20:06:18.000000000","submitted":"2026-05-13 22:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":300,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"912058-bp/dell-powerflex-nvmeof","meta_rev_id":"bb6781633b426af320315b39e225fba149031e38","_number":912058,"virtual_id_number":912058,"owner":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-13 22:52:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":2,"date":"2026-05-13 13:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":2,"date":"2026-05-13 20:53:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":1,"date":"2026-05-13 20:53:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":1,"date":"2026-05-13 13:45:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":37822,"name":"Atanu Sikder","email":"atanu_sikder@dell.com"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 04:54:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 05:00:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-14 05:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 05:29:34.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-14 05:52:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 06:40:03.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 06:42:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-14 06:46:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 08:37:54.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-14 08:59:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 22:00:10.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-15 06:08:44.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2024-03-15 06:42:09.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-15 07:22:14.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-15 07:29:25.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 07:32:23.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-03-15 11:07:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-17 06:46:24.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-17 13:57:25.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 08:58:27.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"CC"},{"updated":"2024-03-22 09:27:11.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 13:37:35.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-03-27 17:01:38.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-04-11 13:43:31.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-04-16 06:04:23.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-04-16 06:04:23.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-08-01 03:45:17.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2024-08-07 14:36:10.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"REVIEWER"},{"updated":"2024-08-19 09:52:24.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-19 10:24:46.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-30 19:05:38.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-08-30 22:39:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-09-09 17:42:34.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-12-06 09:45:40.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-12-16 13:17:37.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2025-01-27 18:23:33.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-03-07 08:05:52.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2025-05-13 08:48:29.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"CC"},{"updated":"2025-05-14 05:08:12.000000000","updated_by":{"_account_id":37822,"name":"Atanu Sikder","email":"atanu_sikder@dell.com"},"reviewer":{"_account_id":37822,"name":"Atanu Sikder","email":"atanu_sikder@dell.com"},"state":"CC"},{"updated":"2025-08-25 10:02:48.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2025-08-25 10:19:23.000000000","updated_by":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"REVIEWER"},{"updated":"2025-08-26 21:29:32.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-09-30 07:18:26.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-20 07:03:58.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-11-07 16:14:06.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2026-01-08 08:53:57.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-01-09 09:02:05.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-03-20 04:55:05.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-03-20 06:04:18.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-03-20 11:16:26.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-20 22:31:39.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-03-23 13:31:16.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-03-24 05:25:17.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"REVIEWER"},{"updated":"2026-05-12 07:12:59.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-12 07:22:47.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-12 13:18:49.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"f02a8b45ab926ba58cc37ff86ab364e0143d7b0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-14 04:54:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39e29c356942cbbd1048aeaed5a60584dc34dfb6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-14 04:54:41.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16635/","accounts_in_message":[],"_revision_number":1},{"id":"fc39315bfcb1238b6528a77e5524dd15c8d43edd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-14 04:54:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16635/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"92c77f8a5d370805009de749b0547fb8177d808c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-14 05:00:52.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/16aa8a76ed9b4c88ba851edd57b5add5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ba34b66361f14004a8efd88d8fd6a4ef : NODE_FAILURE Node request 200-0000002763 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"670cfc060555a6d3a3d1f75689a3d6c8cc698ebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 05:12:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a176a3ce06a34dc496184918d9b3d3a4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e62412306c04f79a8651081536cd4de : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c30643162d9543e8a4a0fd084c661c96 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/abaf0c2dc0c647289b18f6b74bb0c559 : SUCCESS in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ea330ac7a69937e15d69aefca6e5b8259f5f3b9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-14 05:29:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"33ec1a9e3df4586cadec387d0f517b1f0707f901","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-14 05:52:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5b9c4e2dd35a069aee608e24aadaf9797aebe4d3","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-14 06:40:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/912058/1/ : SUCCESS in 1h 36m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7f43062e3acb477938430d150dfc775e0fdaf07b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-14 06:42:20.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/1/check/DellEMC_Unity_FC/d7b4514/DellEMC_Unity_FC/1498 : SUCCESS in 1h 24m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/1/check/DellEMC_Unity_iSCSI/bef8130/DellEMC_Unity_iSCSI/1530 : SUCCESS in 1h 47m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"bafc0a2dfe5e4b731a00618ab7472c6e52e1b455","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-14 06:46:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/1/check/DellEMC_PowerFlex_3.6/2172e38/DellEMC_PowerFlex_3.6/1639 : SUCCESS in 1h 50m 07s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/1/check/DellEMC_PowerFlex_4.5/dd98c12/DellEMC_PowerFlex_4.5/578 : SUCCESS in 1h 51m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc03a059f20ca87947c08fcf9906593a66305610","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-14 06:48:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"030f7f035feebe4f7645a53872b1eb0ca3aa5375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-14 07:04:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"330a3a7df6c73f13ff670c5db19d00297aa34dd0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-14 07:05:03.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16636/","accounts_in_message":[],"_revision_number":2},{"id":"96b4cc9f0dd6db561901f4f77a26a14ef5b195a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-14 07:05:05.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16636/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0dc8155961113b95fd7a7628780f31aa4f37fcfe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-14 07:11:05.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6fcf4664cba443b9a0d6cb02e5daf22e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/25238faf49b8402994e9f68adf995afe : NODE_FAILURE Node request 200-0000002765 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"0d24e5a17ae8989d9b04a3d623d7f8e48efcbaa5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 07:16:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/902d42b1a32646eab99af6b31fdbc51a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e15e3603048448088f666812e8d97885 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd551eabe86a4b30a2708dc6289c39da : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47d9f76e22874a2485f020fbce6ca441 : SUCCESS in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c8690dbe8efcec7392f2442f25417c4b93d8935","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-14 08:02:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":2},{"id":"df85110c09cd79104094085afcc41dd1238f6087","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-14 08:21:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 14m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f2554349b98fd2de8ee7665e35c2afd7232a7f17","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-14 08:33:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/912058/2/ : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1b7f18dcfc8a575f38b7720f2ab8ae424c6435f4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-14 08:37:54.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-1/FC/ : SUCCESS in 2h 40m 55s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-1/iSCSI/ : SUCCESS in 3h 43m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"43756a72ef61409c022d16ef25db819b0a64ce64","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-14 08:42:36.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_Unity_FC/98a94c6/DellEMC_Unity_FC/1499 : SUCCESS in 1h 25m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_Unity_iSCSI/11484cf/DellEMC_Unity_iSCSI/1531 : SUCCESS in 1h 37m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7362f2ea6105cd882a5a7bb11f230139d0ccee0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 08:59:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/168287907bc645cc801860e3163086dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036fa9c92ffd4b7c9e8cea8c0838b0a9 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9845f05523f417f9957999b6799f019 : SUCCESS in 11m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/533ef9e180d847bf83b168e6183d254f : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d375cdfaf744ed8a30f46ab2fd41c61 : SUCCESS in 13m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/a714395094474b81bdcb81b2fbaed7e3 : SUCCESS in 55m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3c3e4dabf4bd4fe49f3a9e6cdf80dd46 : SUCCESS in 56m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72b8f1075640433488d2a887a73465be : SUCCESS in 1h 26m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4198e3eededb4da9bafe10485d7c7ba1 : SUCCESS in 1h 02m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7560c3ec4a9b4768a3a59a21a4967853 : FAILURE in 12m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/632289b55856417e844ac43d6196774c : SUCCESS in 15m 38s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/49b9cfb564b1411fb9f5abdfe9b39456 : SUCCESS in 13m 07s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f12734a1b80048128177faec517fc76b : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03f0d3ad902749e39c2b7d8bc342123d : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4139f72866b746aaae52136fb23de542 : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4df461a93bce47918cbfcabd5524b0d3 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a25974ddf21946f2b1d285fe79cb5f83 : SUCCESS in 45m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/43baf6303d7d4eaaa8591d8d02880432 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/52376e0bfe094e85a652790089837c49 : SUCCESS in 1h 28m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1dd297f6812c4e7eb36f0b8696895876 : FAILURE in 1h 41m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d53f29d9767d442d85d184ea0a7191ba : SUCCESS in 1h 17m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9a0f04af185842a2aba88a671bbc4686 : SUCCESS in 1h 53m 28s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5491bd2d9bc046758e0f6fe1b8673cfc : SUCCESS in 56m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/12b4fe5cf3344df3920f0b57d59df0ae : SUCCESS in 1h 26m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bc65cc7fa28749e69da681000870bf4c : SUCCESS in 39m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f8d89a1532ae47319308c5cd38d22fb1 : SUCCESS in 1h 00m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dd2f92113f2b45a1b1f24733bf73d210 : SUCCESS in 1h 26m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86631e2242c646ffad9e414f0f6005af : SUCCESS in 1h 08m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc82b4e553ea491ab244dd2c6d1059d5 : SUCCESS in 44m 52s","accounts_in_message":[],"_revision_number":2},{"id":"2b46c0da251f1bba3cc6a709a0208f09558d17ce","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-14 09:09:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerFlex_3.6/f08cb24/DellEMC_PowerFlex_3.6/1640 : SUCCESS in 2h 03m 16s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerFlex_4.5/91d095f/DellEMC_PowerFlex_4.5/579 : SUCCESS in 2h 04m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"7f1432a498f6519001298b15abc35d9b89a88ce4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-14 09:40:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"3c77f293f262bc4c346e02efe3dae2b398a53215","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-14 22:00:10.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/2/fc : SUCCESS in 0h 51m 42s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/2/iscsi : SUCCESS in 1h 07m 43s","accounts_in_message":[],"_revision_number":2},{"id":"39e59bf0d606998b7e897136c7c4eca6e39e3d22","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-15 03:19:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"221f9922e574b5b9f39d2b6ef3354b660a04b9f3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 03:26:52.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/48c696db50a542529d39f40ae14de691\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/95e6db33dce14ed8bd135f961814fb5a : NODE_FAILURE Node request 200-0000002780 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"27aa2a39dccbbbb10985bf21889d5049ad430fb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 03:49:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2389fbed48445449352e3e6cd457682\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5022b37f20434266b9b310b68ebf4187 : FAILURE in 21m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2fce98acbe4c46a9bec8f13a7429d219 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c54bd85076344ddd9b449e4de488fc17 : SUCCESS in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c528202971042965d1efceba62a9d8c58cbb9f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 05:48:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00307c837ac1459ca22541ca18269acb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff544f06396a4d1ea7b5ac0138de25fc : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adb518759fa94c39b933ab3476fd67a3 : SUCCESS in 13m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/846d80c1ebf34deda0d0d8fdbff6426a : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e040e3f3fb464fa4573d9d3f6b3f25 : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/5350d29748464cdc99b65bd93d6e0ca3 : SUCCESS in 1h 04m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/67c417f800114ce59b27ad3100b1be4f : SUCCESS in 1h 04m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca5d161eba694f549f6320b61da7c77c : SUCCESS in 1h 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30ddc903044b492c945b06869e4b8185 : SUCCESS in 57m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e284a5d1864e4642825f0854807eb704 : SUCCESS in 10m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/378570277d5b45d98184e3ce796796a5 : SUCCESS in 17m 46s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/016a0e62c3a545a583dfcc4129574928 : SUCCESS in 13m 03s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/618cf97b07ae45cd947711ecd19cfd11 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee301cbeef3f47b1b7f5e1f9d6aac0c2 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/308c47e8117c48a28cdd56b9b950df60 : SUCCESS in 6m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/96c9e88914a54d0593da406645c4b2b8 : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2290d18c622e453aab739632e882af1a : SUCCESS in 32m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21105f8f9cb94b71829b4b45cee26c99 : SUCCESS in 7m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/71880f7c1154401a8ae41f43885cd01f : SUCCESS in 2h 05m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7024fe0bc56748638ac646e525945974 : FAILURE in 2h 19m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9f1a79819dae434980c3baeabb959abc : SUCCESS in 1h 38m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6823e295f5a04da4ab591a0ed17faec8 : FAILURE in 2h 06m 34s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/855c5d8ee8b949a4a615d85707c93e91 : SUCCESS in 1h 10m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/16f22f86217343e6b61dc33086fb89ff : SUCCESS in 1h 28m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/628204cbd5d3433eaff0784337d3a5fc : SUCCESS in 43m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f6dcddc4eda043148631089df9c654fc : SUCCESS in 1h 12m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b66799e96e4c4019b0c6ecf504ddc0ae : SUCCESS in 1h 21m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/191351a446f0468a9924efd5644e66d0 : SUCCESS in 1h 32m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/212ea8e63ba14d95964d5eab93aa6a05 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":2},{"id":"90b15b38871341a7984615be7f6d8e65ce3779a4","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-15 06:08:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"354c64e7853c6649be12ac8d9d84f7514209a2c3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-15 06:42:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/2/20240315120517/NEC-V-FC : SUCCESS in 3h 22m 10s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/2/20240315120517/NEC-V-ISCSI : SUCCESS in 3h 36m 48s","accounts_in_message":[],"_revision_number":2},{"id":"cdc33ccab2cbd04abf1946ac2acf43095ce45d0e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-15 07:18:55.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_Unity_FC/4a30da1/DellEMC_Unity_FC/1507 : SUCCESS in 1h 24m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_Unity_iSCSI/ea81900/DellEMC_Unity_iSCSI/1539 : SUCCESS in 1h 30m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"32677f8845effb1f5529944f33b91d67195147f9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-15 07:22:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/2/upstream-check/cinder-tempest-plugin-ontap-fcp/43b0b6b/ : SUCCESS in 1h 04m 57s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/cf1b67a/ : SUCCESS in 44m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/2/upstream-check/cinder-tempest-plugin-ontap-nfs/846a15b/ : SUCCESS in 1h 16m 54s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/983cd10/ : SUCCESS in 41m 29s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"9773db4c14ffc3ea12a31df38ef250bf96c8c60d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-15 07:29:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/58/912058/2/check/fujitsu-eternus-dx-iscsi/6201891 : FAILURE in 3h 16m 08s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/58/912058/2/check/fujitsu-eternus-dx-fc/5171791 : FAILURE in 3h 16m 41s","accounts_in_message":[],"_revision_number":2},{"id":"625b3a6d76e8ad0d976dade30e94a3135096f2f8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-15 07:32:23.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerStore_iSCSI/5fe0469/DellEMC_PowerStore_iSCSI/1161 : SUCCESS in 1h 09m 35s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerStore_FC/e87b494/DellEMC_PowerStore_FC/1331 : SUCCESS in 1h 34m 27s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerStore_NFS/c05ad2f/DellEMC_PowerStore_NFS/849 : SUCCESS in 1h 08m 48s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerStore_NVMeTCP/1c6dc98/DellEMC_PowerStore_NVMeTCP/711 : SUCCESS in 1h 43m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"d2c7343c68e5d3184c0aad3c28e329ce1657265a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-15 08:22:52.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-2/FC/ : SUCCESS in 4h 10m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-2/iSCSI/ : FAILURE in 25h 17m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"3f4c6ab4d09e745b21e1876521d261c956b41ffa","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-15 09:46:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae3f937ab0428ad8a3fb3a4747a174b8de4b2134","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-15 09:48:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a6ab06e3e06fb0745548592b846441bb732a447","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-15 11:07:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/53/.","accounts_in_message":[],"_revision_number":2},{"id":"a7b9fbff13b3912535b80f6668a212824650ac89","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-15 12:41:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerFlex_3.6/8133134/DellEMC_PowerFlex_3.6/1650 : SUCCESS in 1h 43m 15s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerFlex_4.5/0d572fe/DellEMC_PowerFlex_4.5/591 : SUCCESS in 1h 46m 04s\n- DellEMC_PowerFlex_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerFlex_NVMeTCP/5834c03/DellEMC_PowerFlex_NVMeTCP/4 : FAILURE in 2h 52m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"f36c4670bf0571c8c5dc32cb40fdb1bb02cb043a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-17 06:46:24.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerMAX_FC/800bddb/DellEMC_PowerMAX_FC/345 : ABORTED in 10h 30m 01s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/2/check/DellEMC_PowerMAX_iSCSI/c6bed32/DellEMC_PowerMAX_iSCSI/365 : ABORTED in 10h 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"089175b78ff93cea9a8abcf610501063fa55c977","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-17 13:57:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/2/20240317133423/NEC-ISCSI : SUCCESS in 8h 47m 51s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/2/20240317133423/NEC-FC : SUCCESS in 9h 22m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6fb7a2c715133d86cb289540aa57bb037363e78b","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-19 02:57:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"467850c445f44532c2f5c26f13bdb7166bb1a0a6","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-19 03:01:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9b7a5522eee80be6d4ce1f33da006d5d5f1f65f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-22 08:10:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d52c66e11774c5351078f507ee8c04299d0f09b0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-22 08:18:47.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2926d75423d14156b540bfaa25b5f4e0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/822258ffa87840c492b900e27d440757 : NODE_FAILURE Node request 200-0000002866 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"5fe4aef868c81e84ee0c886021d45c54f6c07126","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 08:33:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be894b2528e444fb33f4b9c3d7e925d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5e55b82e284467ba9cc3c3cd462b58b : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb10ec41d94143dd9d160581c3878ce9 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce21674559f54caab517f59235131129 : SUCCESS in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bdce04005f1f79bbf1853057aed509e33e568c7c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-22 08:42:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 29m 47s","accounts_in_message":[],"_revision_number":3},{"id":"2b7a63531f9da098d4ab3ba8c8f88fa6050d537a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-22 08:42:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : FAILURE in 30m 20s","accounts_in_message":[],"_revision_number":3},{"id":"2254baede7438971aa73ee833f1d8f3f913421cf","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2024-03-22 08:58:27.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5ef25e1ae9c766ba22773c9c55750c8bfdb8e7c1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-22 09:11:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 25m 22s","accounts_in_message":[],"_revision_number":3},{"id":"36bcd0aee37ddbc74c59c98fb59fbba3536c56c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-22 09:27:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/46de5b7d54cd47308a3b8f9269c19b7c : SUCCESS in 1h 14m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/29ef370421074ffc8a58b08db9181b00 : SUCCESS in 1h 11m 19s","accounts_in_message":[],"_revision_number":3},{"id":"0cfc87d36c07a9fbeb71950e604d02eec92b6ad3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-22 09:31:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/121/.","accounts_in_message":[],"_revision_number":3},{"id":"abe076f44642a26974102742f9e42f7f63740bf6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-22 09:35:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/912058/3/ : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":3},{"id":"ed0960ab4bb187754188e15259df09596a568751","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-22 10:05:06.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_Unity_FC/2fd1df2/DellEMC_Unity_FC/1573 : FAILURE in 1h 36m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_Unity_iSCSI/0e57da8/DellEMC_Unity_iSCSI/1605 : SUCCESS in 1h 54m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"fb50a94c2ec019ce5af90ed1f2454d61c3b811a8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 10:43:18.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_3.6/d464776/DellEMC_PowerFlex_3.6/1695 : SUCCESS in 1h 46m 38s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_4.5/90142d9/DellEMC_PowerFlex_4.5/632 : FAILURE in 2h 32m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"34da5b7ead76577128e411e40c907aad5fa6f4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 11:15:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df81b4df7a1144f58288451204653ae4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bfa906b9d9e4fa384f07c351d30369b : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6737e1e7160848a0a25f91b30f2a284f : SUCCESS in 9m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/faa04bd668f04bfdb962450214054bad : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/841567eb85da41bc9290daeadd4e30e2 : FAILURE in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7ba0cc952341bd8e35bb88f5e46f66 : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/f2d3bd0db8554dac9906485552318fbf : SUCCESS in 1h 09m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8284396f5a08499d9ca4af16478599e9 : SUCCESS in 58m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8105262483574582a4ecb0ca564047bc : FAILURE in 1h 31m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d651071fd3345a690adb020bb44a9bf : SUCCESS in 1h 04m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04c0b00cb5a44a4099e63d3b4d101ca6 : SUCCESS in 12m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4cd820763577492c98298c3897b61a7f : SUCCESS in 13m 50s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/df349fe36f30400ca0aec5921ea3e750 : SUCCESS in 13m 03s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/92b8130ea4634d32939a1725a2399cad : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8545809f979458e8b6f588242dfa464 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95e9fd5ac8974ab395fd4c1312891fb1 : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4125f8d7823744599ebe5de79491ba70 : SUCCESS in 5m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f1deec2bf42040ca90b136b01fb15c6b : SUCCESS in 44m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97f1d04db2044ae181d83389cc64fb36 : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ccbe9670c70448a9b8f67b75ccabf129 : TIMED_OUT in 3h 03m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d4aa756ef774e27a9ebc71d72fb02ee : SUCCESS in 1h 56m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae847dfa9a9b41dc8862377200888157 : SUCCESS in 1h 25m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e166c8fc1f1248029e434c1a063713a1 : SUCCESS in 1h 38m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bd22ad0f3af44d9bb04b75109bd73dfa : SUCCESS in 23m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3aac3360f99a479a94254af8192c504b : SUCCESS in 58m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fe8b02539bc843b8a52def426fe1ef00 : SUCCESS in 1h 26m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/228b36dc0be44948a448902cedf2d88b : SUCCESS in 41m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6b9d011022844428ab52eab78b4c748a : SUCCESS in 1h 07m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eb391a3224384d179a8b29d2bef9a043 : FAILURE in 1h 02m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46b0cc08f985450db54dff3255232951 : SUCCESS in 1h 32m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c37adf3b050142bba388082540a88493 : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":3},{"id":"bd50d2be700964628ac52ba00fdea0aee92b1e9e","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-22 11:32:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4148dc17ef16535d602ad5deba43b6e56ee95dc5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-22 11:40:21.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6577c248270f44918731afecd33106bf\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8782462c604e4321b5796adec922f11a : NODE_FAILURE Node request 200-0000002867 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"79afa4a3307e5fb75d3a8374700d2a65734b04c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 11:46:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aaca4b08faf44b5937666b174583317\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f644594f1f048ffbf2ab8cb22cc5c0a : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ad8d4d3ba09486288a903fd07139b73 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dd403fbf57474b0886941591e6f31257 : SUCCESS in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa4f1bed9959d97f999e6e6312dbb976be85b3cc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-22 12:46:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e0ab2827dde24178ab1c2206623e1f7b : SUCCESS in 1h 10m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5a1ad8127b794d5b82fe75f110b677a3 : SUCCESS in 1h 12m 06s","accounts_in_message":[],"_revision_number":3},{"id":"4db67d31526d85b51195163c182c40c99d15774b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 13:31:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb985a293014856ad7c8692170fea68\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be0991a3b2ef49c58f0b7771a32d767f : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14ea42f482404dc997b2766de05106ad : SUCCESS in 10m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/457bb62c1e4246ef9f4fffd96f465a51 : SUCCESS in 10m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74ebd84c2f4043d298846e23819db3c8 : FAILURE in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02b12931634c415aa5d35331d067d74a : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/d2ad63058ae44a7ab9338514ab4464b6 : SUCCESS in 1h 02m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a7e27408277146aa8ce1445e420af4a4 : SUCCESS in 59m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b2ca9dc6d2d4592a63b440ef7d5f437 : SUCCESS in 1h 27m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/499648d88d524c5f98ca98dbd5811328 : SUCCESS in 59m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6af8c1e9eb994c63a2c37c3da7809d6b : SUCCESS in 14m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6c1a88a045444f408c9e4d36fea7a4a1 : SUCCESS in 16m 36s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e3b2e6fd9dbf4dd4a0fccdbe09dcb32d : SUCCESS in 14m 19s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/73f6a965bbf244e898e39daf0f646b99 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a90261276804240be2783b683d9595c : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0d3a161006db4a008f73c72e6a1d844c : SUCCESS in 6m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9ace69960de4cebb04abe0223632974 : SUCCESS in 7m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1fcf11e83e0d4847ab3fc297e2b3c3e1 : SUCCESS in 49m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b038bd8f4b44c26b2e80d0b4f0967d3 : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3771edf621704c8a87e902677866946c : SUCCESS in 1h 40m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/be24df7aacdf4c1383eae1c32dfd68fa : SUCCESS in 1h 40m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66c977531d4a47d7812f0c5dc7c45ff6 : SUCCESS in 1h 43m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/055e151c041543c0a95a5edb26f040aa : SUCCESS in 1h 52m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c099482ed85644029c6eca040e15c67b : SUCCESS in 25m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9693b2592e8a4b6ea605ac2597839efd : SUCCESS in 51m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/68db91f9442646a0981ceaf4705fa143 : SUCCESS in 1h 34m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a40944d2f59148f79559aeeaffbbb51c : SUCCESS in 59m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1cf933f8341940a394d4267e0648aef5 : SUCCESS in 58m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9631b441369d4552890e745b716901f2 : SUCCESS in 1h 42m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ba61949309749d08493f54783841bf7 : SUCCESS in 1h 43m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/296118f17c5f4062a20be2c7fe4e9738 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":3},{"id":"bea0fcbe59b9223f15d622469df643def6a05550","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-22 13:37:35.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_03_2024_13_31_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"985691cc231966ef6fe638e8a1080db1d085c66d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-22 13:52:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/3/upstream-check/cinder-tempest-plugin-ontap-fcp/5a61d1c/ : RETRY_LIMIT in 4m 01s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/c0f1c59/ : RETRY_LIMIT in 4m 02s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/3/upstream-check/cinder-tempest-plugin-ontap-nfs/8858796/ : RETRY_LIMIT in 3m 45s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/b5bbd85/ : RETRY_LIMIT in 4m 03s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"365ef1bf61fcee3559d6372838dee4c14e7ebd6d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-22 15:13:33.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_Unity_FC/8984582/DellEMC_Unity_FC/1574 : SUCCESS in 1h 23m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_Unity_iSCSI/ed63bd6/DellEMC_Unity_iSCSI/1606 : SUCCESS in 1h 42m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d2625d1a79b91d54b43ed149cdbd57e6b1109039","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 15:34:18.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-3/FC/ : FAILURE in 7h 23m 20s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-3/iSCSI/ : SUCCESS in 3h 29m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"4cdb5ec5f456c8b55b036954e3b757756829ed8b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-22 15:59:29.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerStore_iSCSI/0ec4d68/DellEMC_PowerStore_iSCSI/1201 : SUCCESS in 1h 14m 35s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerStore_FC/125c1ad/DellEMC_PowerStore_FC/1371 : SUCCESS in 1h 44m 59s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerStore_NFS/dcdd896/DellEMC_PowerStore_NFS/889 : SUCCESS in 59m 48s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerStore_NVMeTCP/998c312/DellEMC_PowerStore_NVMeTCP/751 : SUCCESS in 2h 28m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"d25f47331dc5e97171eeef9c20606627e2f0cd3e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 16:04:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_3.6/744df31/DellEMC_PowerFlex_3.6/1696 : SUCCESS in 1h 43m 18s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_4.5/f86ed98/DellEMC_PowerFlex_4.5/633 : SUCCESS in 2h 02m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"32203630a4a77e7a610f57ee96fe48a18e251f26","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 20:14:43.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerMAX_FC/90b6d68/DellEMC_PowerMAX_FC/360 : SUCCESS in 4h 39m 51s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerMAX_iSCSI/59ce644/DellEMC_PowerMAX_iSCSI/380 : FAILURE in 5h 29m 15s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"a533ba60d7ec6a9e6041964a60e51e51e5f3c316","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-25 03:48:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"374340c67344a1f09eb26d99a9c8bf38fa899820","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-25 03:51:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c433942ec0b65c948c309aae7c30ccd3d649700a","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-25 04:12:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11717d70fbbda23048d837adeee0f3d43217df6d","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-25 04:15:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7a864562912f131d6a7b313e7dbcbe33880068f9","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-25 04:18:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fc4273164d20b9c4a8d3a6009b39ce0dcb6f19b","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-25 05:03:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6761062d9d4365c772a91c6e2ace67113cca40a4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-25 06:00:11.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/3/fc : SUCCESS in 0h 31m 59s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/3/iscsi : SUCCESS in 0h 45m 39s","accounts_in_message":[],"_revision_number":3},{"id":"c967810649821db8a9ca6dc140f11689e6ffdfc1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-25 06:59:27.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_3.6/030d10f/DellEMC_PowerFlex_3.6/1088 : SUCCESS in 1h 33m 27s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_4.5/089b838/DellEMC_PowerFlex_4.5/65 : FAILURE in 1h 54m 50s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/bc05dea/DellEMC_PowerFlex_4.x_NVMeTCP/2 : FAILURE in 1h 55m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"a93ba43f6625d8f6a23b835d90177c747b177cd9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-25 09:27:21.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"08058ca5fc93641e5b4e504919552bee7b0666e6","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-26 06:09:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17f3301b44471cce6ce23039c4a70795124bdf61","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-26 11:32:59.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"ed7b90ab3f5eebe50118cc69f0b0e5fb1e886c4e","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-26 11:34:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e08726d11a1fcb7cfa74f1f16e728fe54a949b0a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-26 13:40:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_3.6/29b1c66/DellEMC_PowerFlex_3.6/1093 : SUCCESS in 1h 47m 39s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_4.5/a2c6245/DellEMC_PowerFlex_4.5/71 : FAILURE in 2h 05m 44s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/a066691/DellEMC_PowerFlex_4.x_NVMeTCP/7 : FAILURE in 1h 26m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"e3477886b33fe4739d4c5dfd6fa35b2f18ea672a","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-27 02:18:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"52b65bac4139cb04dc3862699a3482c2e7c721d2","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-27 17:01:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/912058-2/Lenovo-iSCSI/ : SUCCESS in 2:05:04\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/912058-2/Lenovo-FC/ : FAILURE in 3:02:21","accounts_in_message":[],"_revision_number":2},{"id":"25d6d9efd5752d1e222e6175babc028befe133c9","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-03-29 03:10:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2d0fa24e6cb4aaf4124376d36541f7d87dd4ecc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-02 04:58:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b97a17e8958c4a03964785f0040e50230c26a2c2","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-02 05:00:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a89231f02a6aca0f7c2eb8a139e4bc042ca328f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-02 05:05:50.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/02ad330c8720445ca89d9e3279f7e83c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5615f0eb2ce54f1a82ed811fc0814925 : NODE_FAILURE Node request 200-0000002916 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"5f5c23bbad7fd03da47214ac1e20ffb9984746e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 05:10:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00eb4b1f8aa34644baaff279c3312af3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e330f45ec9f4239bfef8e3f2aa1d185 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7d26c535a3e640dfbd66abc4fd8e1a1c : SUCCESS in 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f8bcb29aafc30d70015aecb380f56debddb856d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-02 06:22:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/157/.","accounts_in_message":[],"_revision_number":4},{"id":"4f77ae10112aebffb0a35f48cbcfbd9a679e5cc6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-02 06:32:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : FAILURE in 1h 32m 17s","accounts_in_message":[],"_revision_number":4},{"id":"117dd97caf817271eacfac512ec19bfc1a7fb72b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-02 06:32:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 30m 34s","accounts_in_message":[],"_revision_number":4},{"id":"87ebb3cb5a07ee5b69a9cb0d1d24dd47def1fe52","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-02 06:43:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/4/check/DellEMC_PowerFlex_3.6/278b32a/DellEMC_PowerFlex_3.6/1121 : SUCCESS in 1h 44m 08s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/4/check/DellEMC_PowerFlex_4.5/240f4d7/DellEMC_PowerFlex_4.5/99 : FAILURE in 43m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"0687faaeecaf3a6c66580ac2a38499d595ddc279","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-02 06:49:02.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/4/check/DellEMC_Unity_FC/2aa2d73/DellEMC_Unity_FC/1615 : SUCCESS in 1h 27m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/4/check/DellEMC_Unity_iSCSI/73760db/DellEMC_Unity_iSCSI/1646 : SUCCESS in 1h 50m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"bc014c459aee2d2cdb85d497b78329e547b7094d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 07:04:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcdf01a161734a1f9ebb39653d2d5af7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95084c173e14c1f98aa7a229a47a3d1 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8576cf2c5494cc8a7b490bc54ec23b0 : SUCCESS in 11m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7e215069093427aa8066a1c50460308 : SUCCESS in 8m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69426ce46bbe462cb04f30a00a882f89 : FAILURE in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc6d8ae1eb8440c1ac2c9ecb27fc7925 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/0bb9fec3e68940f1aaeabe5c1af1e122 : SUCCESS in 56m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f629022bf14c41bd90bbf0f89b1c6e2d : FAILURE in 1h 20m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d0aacba96af4b4c979959c9062c941e : SUCCESS in 40m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d843677679441ba9e3be029a37c62ac : SUCCESS in 10m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fb538c7a91054d909cdc410e5e426c9b : SUCCESS in 16m 42s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bd7d6aef2d3e41beacadbf0f616c9afa : SUCCESS in 12m 36s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ed7cd1f21bd54b00afed871246659ff2 : SUCCESS in 4m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6df48d9cfa6747198dbc9ed581a9cf61 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3256e753cf6a40509c9b55eee60987fa : SUCCESS in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7235a8b6dbca428593d0f18d4e339e03 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9f27623551414217b188ccac7dec450d : SUCCESS in 31m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f8ca015e421a4fea88a88812e19aae8f : SUCCESS in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d59f4d23e39e46b898b5685f96130f35 : FAILURE in 1h 21m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2807413ad26d44eb94177e2f044f5d20 : FAILURE in 1h 39m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ff4d8304d69e447f8a23b75a48ebf266 : SUCCESS in 1h 45m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8ffe87b443674853aabe31d7ce3dc3d2 : FAILURE in 23m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/da9cff91de604c6cb1e049ad4aed07e6 : SUCCESS in 25m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b369ded17a4c45b5a1378be386695d8d : SUCCESS in 1h 09m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d278d593418c41fb867d123d04898221 : SUCCESS in 1h 29m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/65b8973c3f364e0f93b1bdde1acc1427 : SUCCESS in 27m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/60f99edfc0c4445fa198524610e53107 : SUCCESS in 1h 25m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1963d48d79874c318902b959760e76b7 : FAILURE in 24m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/abdfa1a0f2aa46619ffe09586b76af14 : SUCCESS in 1h 46m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25d535a0ff8a4bc89ee86e4b0d5cc3ae : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":4},{"id":"681ab380bc53152f713d1c789f3fc887bdf7ca72","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-02 07:10:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a62488d8b60f579c5012886c09393a3ee10b7033","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 07:33:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/215f9c70303c4309a469713c2bd9b58e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6b4481a40b54f7a98378c147f971464 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9a1e01f42be43dc907b6f6f528edef4 : SUCCESS in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a97050f9cd16cfb86303f5ac71dedbc89a89c7cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-02 07:45:04.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ca45e11afe3d429ab93f0b681067734a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d1cdf5adff734de987c67123e9044197 : NODE_FAILURE Node request 200-0000002917 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"f995db547ebfc128aa8ef6f0dd2db51e981ca780","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-02 08:15:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 41m 06s","accounts_in_message":[],"_revision_number":4},{"id":"e4793947ce940c043895af46c0b32b2356910f7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 10:24:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ca005d5338a4651a704bc9108874e5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69af27227bfe4cab9e933d8ffc1eceb3 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78859fe52cec480a9afcad82ef13ed62 : SUCCESS in 11m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7cc0306d5b7d470d8413d9e8288f1bba : SUCCESS in 8m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ad1597be79b4f9983e6e0f8383736dc : FAILURE in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60008a31125f4088a58d9d4db5f379af : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/8946cc288961485ea8b8555c22454a50 : SUCCESS in 1h 01m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0daabb084cee45c5a0c42ff0a15eeaf4 : FAILURE in 1h 04m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d6e0d9bc347403fb20a26f6ff50e0b1 : SUCCESS in 45m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f937075eec5e41cf947ce4c1a40b370f : SUCCESS in 11m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c984024588c49c984b8c54d7f5d75a7 : SUCCESS in 12m 50s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dca2057feb3a4350b160bcce7760c7bf : SUCCESS in 11m 07s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da843849d8994daa951b4b90470b4565 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3ab2714945674c56a0256c700b09cf40 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a86812f8f84240999ae4f7cce9f309df : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/56f66054fde042779708b7ddf94c4a08 : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2072ca1369434676b0ca4acc98c22bd7 : SUCCESS in 32m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7084125f945f4ea7a44ab67d5f6a89ff : SUCCESS in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b78f2065576240288f8add01cfaeca7b : FAILURE in 1h 55m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ad881530ec9844bea21af469f7454850 : FAILURE in 2h 16m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0fd6336aae0549e69c886695a2fa70c1 : SUCCESS in 1h 28m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1727e843e61643cb8c141666400b08e8 : SUCCESS in 2h 11m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d710c1bc65f7487bbd2b3f124b6a821b : SUCCESS in 25m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/92bbb89facaa4dd2909392f9488adce7 : SUCCESS in 1h 16m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ef03a823a9994d2881f163f4bbcf4438 : FAILURE in 1h 13m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bb47a4dc65ad4ca09cf9cfb59441c767 : SUCCESS in 31m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1801bf2ff6e84fd69cf3921d952e2181 : SUCCESS in 1h 35m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/134ee4014742429093a0697c4f50817a : FAILURE in 7m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b9f0d8a00c94b5e98fab576530e533c : SUCCESS in 1h 25m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86604acf1fa24fd5b8299945bad4d5c4 : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":4},{"id":"70c3c5ae8c190e71754200fa821dbac6e8d1f10b","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-02 11:35:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eef8bb1701e3f5507b33928fb52c477256405947","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-02 11:43:20.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4e355d237f7d4170935d3d9383a84cc9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/beb6f1b22b554d1b9ab8ca27ae2da160 : NODE_FAILURE Node request 200-0000002918 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"eccd8a1542f9795163d89e243598655b8c3889b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 12:02:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b20ffb857704ebdace1da7670a5d27e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20a009e43d314b5aa9c5344e671cac0e : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/769ac9ac71fb4c9780b688d974529859 : SUCCESS in 17m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"858d27b50fb9d2a0d3c76b448da7956b3d297757","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-02 13:14:55.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-4/FC/ : FAILURE in 7h 59m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-4/iSCSI/ : SUCCESS in 2h 9m 7s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"2d542fed76fb81a9875a401e25bc7780f9aead9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 14:04:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32cfd68157ee426db61678bc77a10d83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2186388c57ac4f839b379346e8ab19a9 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/875a49dfc3404be89cf5474bb8c5de59 : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33bf184eb53c48f1b0e12848a448a8a9 : SUCCESS in 8m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eadd7115b6cc4a53bb914a04bf4a797a : FAILURE in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f1e56a02944d47b6b02b80af62f62a : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/026c6dfa2aff4e05a1567f2b4636f8af : SUCCESS in 1h 12m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d008da3f86cc4702a21acfac263fbdce : SUCCESS in 1h 34m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f7be9a7ef584502a4876fda551f609d : FAILURE in 1h 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a444dad59df45c5afa24550d3fcdd18 : SUCCESS in 13m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/97bcc3dba05b4589b827953545e5edbc : SUCCESS in 17m 18s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0b606e6fd8bc4137ac3456ee5f4c53e9 : SUCCESS in 16m 42s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0bb837438a6f4744b3512251d8f464f2 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2378e750d95645de930db5420bc1ec1e : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fd778b37f5af491785cedb5f4edde49a : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/728f426b3a7e4ff0868deced6e27f0f0 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f29631b264be470f8f7736b66d79adf6 : SUCCESS in 45m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e248bf9fcea4a07b777866b13cd0c0c : SUCCESS in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f8b7eebb8c8b4004b6698aba776a7517 : SUCCESS in 2h 09m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5a1903dff0104cf79d202e0ae71b6caa : FAILURE in 2h 12m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8ddf9e24cd8f446fbc603eee1691fdc5 : SUCCESS in 1h 09m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d9c471aa45214ee688839a2d3fd9b675 : SUCCESS in 1h 41m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/44b095a9415441ee92b096017841ece5 : SUCCESS in 37m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5e541931d6d548ff8bd60e0271d46803 : SUCCESS in 1h 16m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0dbb52b93e5648f5a1e44aa4e9d8a4e3 : SUCCESS in 1h 25m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/18950e7680f649d4b07af3c375cc3d35 : SUCCESS in 30m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e508bef2d0714c06aea77765f5e02a4a : SUCCESS in 1h 12m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/904210c38e2141cd9632171e3f11d3ad : SUCCESS in 1h 57m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e15eb0b50cb4b49901f552b5d20038d : SUCCESS in 1h 10m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74236b3294a84837baea0302460aa8fa : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":4},{"id":"7dcbd75a537c3fff81e6e1c86b876913e96bf43a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-02 14:41:44.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/58/912058/4/check/fujitsu-eternus-dx-iscsi/f0ea5ff : SUCCESS in 2h 56m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/58/912058/4/check/fujitsu-eternus-dx-fc/2d682cf : SUCCESS in 3h 02m 54s","accounts_in_message":[],"_revision_number":4},{"id":"c7dffff0e44b8cdf9ddb11b02ff87d268f0906f2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-02 20:40:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/4/20240403005338/NEC-ISCSI : SUCCESS in 4h 36m 20s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/4/20240403005338/NEC-FC : SUCCESS in 4h 46m 35s","accounts_in_message":[],"_revision_number":4},{"id":"c42ad5c0926f530eba60a6c8116fbc583ff98d30","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-03 02:17:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b93cc0d440f8921ce9d49079018689c6e77c9ed0","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-03 07:31:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ee92020f6af4d30516c9699c8e5037bc3ee8eb5f","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-03 11:28:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36797342c9a97a9c64422d08e86f11f912c4bf42","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-03 11:29:05.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/4/check/DellEMC_PowerFlex_3.6/2955944/DellEMC_PowerFlex_3.6/1128 : FAILURE in 3h 57m 20s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/4/check/DellEMC_PowerFlex_4.5/6e2bfdf/DellEMC_PowerFlex_4.5/104 : UNSTABLE in 2h 37m 20s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/9100617/DellEMC_PowerFlex_4.x_NVMeTCP/25 : FAILURE in 3h 32m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"dd33902436087bb7e3c16e301a19ada172796174","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-03 11:36:33.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fb96ec9d6a8c4a2a9c9dd8a854584ae2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/909171b2f02a4f1b84674d9b3d894ad6 : NODE_FAILURE Node request 200-0000002924 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"e8c42263fa08e79993b177279fb112bdd1480c54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 11:50:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18130ab686b544cba00c1784a3b12a6a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33816fbeea014a9bbff82633f96fc458 : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c63ffb6ca834a708afe1520e8c38732 : SUCCESS in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"46efa79fe9385f40bc119f106706140a3bd380ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 15:03:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6cfea1d93414d89a72d96fafd615fef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c67fea375645e1a69772b3da100893 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd3e8700b5b0459d974e905ce14fa6b1 : TIMED_OUT in 41m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e67b376140664367999f7c8918394a41 : SUCCESS in 12m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6629154d6c714881aa0712025188657f : FAILURE in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455886178c564dfe8f47361ef98c0980 : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/edc67bbdb94a4d1c89afcfa4bc2bf629 : SUCCESS in 1h 13m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a62f75a34a484679969bebceec7da9b5 : SUCCESS in 1h 22m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7dd9bd588bb410ea98d005a719aef0c : SUCCESS in 57m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f9bc7d2c3d544cb9bf41fc804fee7f8 : SUCCESS in 10m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28e6af67fd8b4d7c9574966d87c4c98e : SUCCESS in 13m 57s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/99f669c0323f4fdda4a4ad02e33cce14 : SUCCESS in 12m 40s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f6d208f6a49e45f6b59cde01d9602d38 : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8230bf43f6845f9a28206919fcfd926 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fa0b51e2398e44589569fee30f20d28a : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76761d01cf33433f9323a2fbe0687032 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1b3e313beeff4566ae065dcb340f2172 : SUCCESS in 32m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f7d7a1345574ccc961467a9004716bc : SUCCESS in 5m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5676177d98a34da1aba8ee3ad02dc6b3 : SUCCESS in 2h 03m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c27a1094ea4c48d89c5c6a8d4d17f17c : FAILURE in 2h 18m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2a6a2f76c421432fa437dc4a5bb919b1 : SUCCESS in 1h 44m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/742415aed4c34affa6dfd386d73a5358 : SUCCESS in 2h 38m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/754c773f5c334ef8ae092e601f94c3f0 : SUCCESS in 47m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8d6eb12ccf564332b2c4a1b2d2107847 : SUCCESS in 1h 21m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a168392f7b8e4206b4a085b72628959f : SUCCESS in 1h 08m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3884db619d494957b5bcc010bcdbc7db : SUCCESS in 41m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6404fe46cf33475aa305fd9e0ce60a7a : FAILURE in 18m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/92fe017489bf48bca10ab5cf1f540a20 : SUCCESS in 1h 33m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55a26c5920284f99972e9855f30fc810 : SUCCESS in 1h 47m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eb46197655f42569896aaf41f9019f7 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":4},{"id":"0e2496ad1f14c9d0aa478b79bbd1c4d97f9a26ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-07 06:19:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"dd8aeee10a3e9a4f6ac89940bf30e30af7d7a646","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-07 06:34:48.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1ef4964a8f0c49d6964177ca3f330c33\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/519c2fb50ec647d78f975a3445167e2c : NODE_FAILURE Node request 200-0000002947 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"7f8b1e80dc6fdaf76700d4c0fdb2fbeccb0b1ca4","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-07 06:43:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ea44ce72fec7695a0c96d04848455f3abc5cd9f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-07 07:49:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/183/.","accounts_in_message":[],"_revision_number":5},{"id":"7935aed9a9eaf862879a18bb39282fbe3060d479","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-07 08:06:59.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/5/check/DellEMC_Unity_FC/9b8ef64/DellEMC_Unity_FC/1641 : SUCCESS in 1h 21m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/5/check/DellEMC_Unity_iSCSI/dcf8fca/DellEMC_Unity_iSCSI/1672 : SUCCESS in 1h 47m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"2226483c3064761db7101a98e550cf57b667d0d1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-07 08:49:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/5/check/DellEMC_PowerFlex_3.6/80d4ec1/DellEMC_PowerFlex_3.6/1143 : SUCCESS in 1h 38m 54s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/5/check/DellEMC_PowerFlex_4.5/3f5cfbe/DellEMC_PowerFlex_4.5/119 : SUCCESS in 2h 01m 22s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/4c2a026/DellEMC_PowerFlex_4.x_NVMeTCP/28 : FAILURE in 2h 05m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"59f7f554b829485fde8f9bec6ce4a8d3826e98ac","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-07 09:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/5/fc : SUCCESS in 0h 40m 31s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/5/iscsi : SUCCESS in 0h 48m 42s","accounts_in_message":[],"_revision_number":5},{"id":"256a20e7d9af2fa665949774b1edfc20908c8898","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-07 13:39:31.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-5/FC/ : FAILURE in 7h 19m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-5/iSCSI/ : SUCCESS in 2h 45m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"ba0028e3ff1f6d24d29e2073785453f384b440af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-07 13:56:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e2bb0613c2f98a91b1d3a972721f2d927c91ab0a","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-07 13:56:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f18284256a26cc9c8c820265ef60bfaf4ac5c965","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-07 14:02:57.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5524696eea4e4638a6f829702f5e303d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5ab7ac5d584f428fbf1a5c4a4ca05611 : NODE_FAILURE Node request 200-0000002948 failed in 0s","accounts_in_message":[],"_revision_number":6},{"id":"f6401a6ae84ee173c85f08ae767c784210f89fb7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-07 15:23:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/184/.","accounts_in_message":[],"_revision_number":6},{"id":"222b08109e5e376c5b506c9a1a1746a9b1236c3e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-07 15:40:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_Unity_FC/ed14d42/DellEMC_Unity_FC/1642 : FAILURE in 1h 23m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_Unity_iSCSI/2673d6a/DellEMC_Unity_iSCSI/1673 : SUCCESS in 1h 43m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"5bb7c18ae5f8c7c027b323d41dbce30ff0ee3aaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-07 16:50:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d1fef043a64a4e91f78778df840095\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9595ab343804424adf114d65871faf5 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a73288b6f3c46c5a14571f1b03c1889 : FAILURE in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b7a1e95352fc0b8af04282e80782632019d8c061","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-07 17:00:07.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/6/fc : SUCCESS in 0h 31m 24s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/6/iscsi : SUCCESS in 0h 49m 46s","accounts_in_message":[],"_revision_number":6},{"id":"2d22e24903dfc7dfebb7ef8aaf45602e3248a439","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-07 18:03:06.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_3.6/fd9d03b/DellEMC_PowerFlex_3.6/1144 : SUCCESS in 1h 38m 36s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_4.5/e348373/DellEMC_PowerFlex_4.5/120 : SUCCESS in 1h 51m 20s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/4112bf0/DellEMC_PowerFlex_4.x_NVMeTCP/29 : FAILURE in 2h 10m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"fe09907d31b40a1936cd572384e8e92004e7c0b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-07 18:33:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36cbb5996b2943809a46b248d008cb51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb5d6cd5ef64ec3b5f2e4a36fec58f1 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91a33a5626ed474f98d56330dd36c9cd : SUCCESS in 9m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61ee6c8e6b684ed7ba987d03008a273a : SUCCESS in 8m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/545c76b3010b4a3e85c9df76bbbc4cf9 : FAILURE in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806acf36639243899e307540e9e85307 : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/e274016921b04703b5c4d97733f8803d : SUCCESS in 42m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5879104692254f4b91b29cfd866e05dd : SUCCESS in 55m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9a08f63208f433facddae10d32db393 : SUCCESS in 46m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3eeedb0c87c940aa835c22933b63aebe : SUCCESS in 10m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7b6172193716496e9fcd3dc4fce91d7f : SUCCESS in 15m 39s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f72b4181693140ac978fb9fa76372ac0 : SUCCESS in 11m 04s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fb676664e4984cb6ab505dfffc081d3f : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f3bd76369a6c4d688999a1c0ab88ea06 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5819e059c0614f7b8d2ac30d81fcdd93 : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/552cf12cd57c4ee8b4635e441be8800a : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cbd2676f138c4b85aebb85d0e8c48ccc : SUCCESS in 45m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f9a8240cfa12493db6c92dd65b661162 : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/849c227639014e0db6ee9a9a791639e8 : SUCCESS in 2h 02m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/066fa70d39464489a9b35e741f2da1fa : FAILURE in 2h 01m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/461b2d782172432db8fcbb54427656b4 : SUCCESS in 1h 42m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8ee08bc3685a46eab8a61553f5bf8ae0 : SUCCESS in 1h 58m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/59240a64ea67462da609fef61b6e93e6 : SUCCESS in 40m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1154d2f11d624fdfb0449e20bad7d061 : SUCCESS in 1h 16m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eb5b6117cd9e402cbd039ab5bd50d8a5 : SUCCESS in 1h 24m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/52c1a1830e7e45dca0925b396b0d9910 : SUCCESS in 31m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/86a967c9e2e84f5a83b387baf8e57da5 : SUCCESS in 57m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ed7d3dddf25c4457a1495c4e5c753389 : SUCCESS in 1h 00m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7b45897eb19498087ebbfe74988edf8 : SUCCESS in 1h 36m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0eee9219c3d46ed9acd2c71acc32c38 : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":6},{"id":"b87aaff4f2f87ccc94e501414859230044f5221a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-07 18:39:31.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_04_2024_18_33_11\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"f00a9e13795ce80512f7f8a489e7eb71e9bf2686","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-07 19:09:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/6/upstream-check/cinder-tempest-plugin-ontap-fcp/1f18952/ : RETRY_LIMIT in 4m 05s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/dd03b63/ : RETRY_LIMIT in 3m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/6/upstream-check/cinder-tempest-plugin-ontap-nfs/d4cbedf/ : RETRY_LIMIT in 3m 25s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e629423/ : RETRY_LIMIT in 3m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"2eb94f27975966b7165cd6a1a7f024863de96a85","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-07 19:54:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/5/20240407200835/NEC-ISCSI : SUCCESS in 4h 14m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/5/20240407200835/NEC-FC : SUCCESS in 8h 45m 44s","accounts_in_message":[],"_revision_number":5},{"id":"c65fc3063cdbb09536797bf161c4b3ee0df2faa3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-07 20:16:48.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_Unity_FC/2e08860/DellEMC_Unity_FC/1643 : SUCCESS in 1h 27m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_Unity_iSCSI/be64931/DellEMC_Unity_iSCSI/1674 : SUCCESS in 1h 43m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"d8cbe574bd06a4161a38b04a41bf3fe3b06ff337","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-07 20:23:02.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerStore_iSCSI/7f7ecdc/DellEMC_PowerStore_iSCSI/1227 : SUCCESS in 1h 17m 21s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerStore_FC/4bda9d0/DellEMC_PowerStore_FC/1398 : SUCCESS in 58m 44s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerStore_NFS/fc1a109/DellEMC_PowerStore_NFS/917 : SUCCESS in 1h 08m 56s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerStore_NVMeTCP/318303f/DellEMC_PowerStore_NVMeTCP/778 : SUCCESS in 1h 49m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"3d9dbceb5bf82bb86157702253593fa644202ca0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-07 20:56:14.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_3.6/73f925f/DellEMC_PowerFlex_3.6/1145 : SUCCESS in 1h 45m 24s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_4.5/34eab61/DellEMC_PowerFlex_4.5/121 : FAILURE in 2h 22m 56s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/fc4046b/DellEMC_PowerFlex_4.x_NVMeTCP/30 : FAILURE in 2h 15m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"225f3643ca090b2b4354b66b53dfa975ca5da3bc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-07 21:36:25.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-6/FC/ : FAILURE in 7h 39m 55s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-6/iSCSI/ : SUCCESS in 3h 12m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"0f2fbd2503c70a8b0c0b8bef515693f86d9cf04f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-08 00:38:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/6/20240408045425/NEC-ISCSI : SUCCESS in 4h 18m 18s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/6/20240408045425/NEC-FC : SUCCESS in 4h 43m 46s","accounts_in_message":[],"_revision_number":6},{"id":"8b6d97a1f176f9e5e6ebb6ca9637a352deb1b00c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-04-08 05:03:21.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerMAX_FC/89eb31e/DellEMC_PowerMAX_FC/382 : SUCCESS in 4h 54m 57s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerMAX_iSCSI/8c09bce/DellEMC_PowerMAX_iSCSI/403 : ABORTED in 10h 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"937651b5e27731568b6dc54fe30f9b9960f9d70b","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-08 09:40:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f88d133d6fe70993125e9dc5ad3db12fcdd9ddbf","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-08 09:44:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"35d734c1c02bd1d385b0de2578113c9aa8a2fe3a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-08 12:13:07.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_3.6/672c6c2/DellEMC_PowerFlex_3.6/1150 : SUCCESS in 1h 45m 26s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_4.5/f261a52/DellEMC_PowerFlex_4.5/125 : FAILURE in 2h 08m 59s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/6a9e3e6/DellEMC_PowerFlex_4.x_NVMeTCP/32 : FAILURE in 2h 28m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"bb99370609358fae9c3b7cbc98377c796cf008fb","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-10 07:45:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0205723667f592dcf19ccd0e8e56a3bd24ca45dd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-10 10:13:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_3.6/f399313/DellEMC_PowerFlex_3.6/1156 : SUCCESS in 1h 42m 58s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_4.5/1fb05f4/DellEMC_PowerFlex_4.5/127 : FAILURE in 2h 14m 54s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/89a2205/DellEMC_PowerFlex_4.x_NVMeTCP/33 : FAILURE in 2h 28m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"b8fb1711f14a968966ad45c57d7ee70b26f2d047","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-04-11 13:43:31.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"62c40ec3c4d429ccafeb979c395c6800922e635d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36952,"name":"Karam Jameel Moore","display_name":"KaramJameelMoore","email":"wawmart@me.com","username":"KaramJameelMoore"},"date":"2024-04-11 13:50:23.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3af8dcc0376b1ec431a2b2ba8ff2e288145abb41","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-11 13:57:18.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5f39624ed9174d66917afe825671179f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/daa124777eb141e5b5cbbc0aa975ba5d : NODE_FAILURE Node request 200-0000002965 failed in 0s","accounts_in_message":[],"_revision_number":7},{"id":"f23674a2e706bedd834bcf97e826c8f7e36a1117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-11 13:59:18.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"b81036e97a27715a92e822895bf8bb3f3dc9d74a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-11 14:06:59.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fb4122a8222e4a7286186ff9a5b7f824\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/113352bec7644c849b75250eb8bb2a5c : NODE_FAILURE Node request 200-0000002966 failed in 0s","accounts_in_message":[],"_revision_number":8},{"id":"86859d8339cbdfb3b83fd73f2f74cd9650b308b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 14:20:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07659ebd58074758b2d24ca993af5533\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/117c16e2f474492f93fcc7968d088128 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b2787079386b46fd9578b2d1ec3f806d : SUCCESS in 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cea798e987f7953ebbd222bac3b68b2c14037af2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-11 14:49:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : FAILURE in 46m 05s","accounts_in_message":[],"_revision_number":8},{"id":"bbbd10fd53b7ce42ce9f1b205a50c659d039dac3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-11 15:44:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/201/.","accounts_in_message":[],"_revision_number":8},{"id":"d4035804d7362a5701bdceba335ffb0ca3b043eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 16:29:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13708dda75874b85aa45e9505f4e55cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1040f941e9494490be95bb222818b30c : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13543dd69acc467599ae8121063b714c : SUCCESS in 13m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbbfeb64f6f74c9095d3a3a372aa2505 : SUCCESS in 9m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdf0ab328e31411991305da345018523 : FAILURE in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f7e0e9e48c492f969df0a482a02285 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/53fc9dca7b3242019031fd3ba3aa79fa : SUCCESS in 48m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9dd09340f4ce4924acbc18513ec4805d : SUCCESS in 1h 20m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c79c57a9966c49ca9da17db340e17215 : SUCCESS in 49m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23dc4a12f6644f32a8355ff9cb109c34 : SUCCESS in 16m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2405cb27d9740dc8ad1de5092d9d9dd : SUCCESS in 16m 14s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/17550d2750974a71be1c5e5ea28fdd50 : SUCCESS in 12m 21s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c7300b148924e5a89cc24e6071b2b69 : SUCCESS in 4m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4db52d87b1eb48c6873ba358ed9ff8f0 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3407090d51bc49a098a2447aedb1d181 : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2bfaab69bb9c4d5f83a821d836095328 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dd79efb76dd5469f96447db607d7351b : SUCCESS in 44m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd4134f4c2d341e3b2ffcd7b4b66b364 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c824f2c05d284070b2a1a9d9f09c6c28 : SUCCESS in 1h 58m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6fc7a916d55d4d38ab67401072380d67 : SUCCESS in 2h 17m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e3f56a3f72c64f1d8d82546635d2db0e : SUCCESS in 1h 51m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/46d7aef73b4342e48291d51bbec956c9 : SUCCESS in 2h 29m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7c2e7e5aebe7463cbfbf2dc44b296b04 : SUCCESS in 35m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/882f841601504e6daf67745370eaa03b : SUCCESS in 59m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a20123e295ee416a84a247d9e0e22bc3 : SUCCESS in 1h 22m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8ac5ab7338ca4831a6aa0b85e4fcdac9 : SUCCESS in 47m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/82bcb28544034256a02036cd118a2feb : SUCCESS in 1h 31m 25s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2414b504cd09409ab6f8e77884737ac9 : SUCCESS in 1h 45m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86c52db202d44f3294799a0f32b56e5a : FAILURE in 1h 32m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae5edf3cede4458a961661065fb5e645 : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":8},{"id":"c4357b74ed4071dc95807799b6e8b20a6c407220","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-11 16:44:04.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_Unity_FC/d82419e/DellEMC_Unity_FC/1656 : FAILURE in 1h 18m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_Unity_iSCSI/99b2db5/DellEMC_Unity_iSCSI/1687 : SUCCESS in 1h 44m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"88d3948178d6d3571fc8050abd0b75c725587c10","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-11 17:00:07.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/7/fc : SUCCESS in 0h 46m 10s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/7/iscsi : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":7},{"id":"c62ca4a6475a6df8e074a62bf9801db51d8752bb","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-11 18:00:08.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/8/fc : SUCCESS in 0h 57m 08s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/8/iscsi : SUCCESS in 0h 53m 00s","accounts_in_message":[],"_revision_number":8},{"id":"b9b551a720102d44dbea5cba2fcb86bfa2122c48","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-11 22:01:19.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-7/FC/ : FAILURE in 8h 10m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-7/iSCSI/ : SUCCESS in 2h 11m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"77dd98ca236a98b1e9c8c87bd87d4cc343a9e958","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-12 02:19:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a570e078e95762859703ca2aae9ad78bbd1edcfe","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-12 02:21:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dfa60aab30e2c668c16aa369c04a29616935fdd3","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-12 03:36:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"658165048d44924db9f9136ec2f9dd282e2330d1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-12 03:42:58.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5473845a87334d4cb5999102f7e3832f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bb8856d8d7784ede8ed571ca8287ad76 : NODE_FAILURE Node request 200-0000002970 failed in 0s","accounts_in_message":[],"_revision_number":8},{"id":"93769e473d223c88e16261cd8399c6c4a140cd3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 03:49:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36fd26ed791c40fd93e87efa36063e07\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d79ac1f6088645078b58906df283bb06 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/461bb4653cbb457a8ab2c9ad8b2e025c : SUCCESS in 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ff7b5dacec78081bbee9c16c22a3e5b12d995f5c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-12 04:25:02.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_3.6/f5a0112/DellEMC_PowerFlex_3.6/1167 : SUCCESS in 1h 44m 18s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.5/337cd0f/DellEMC_PowerFlex_4.5/138 : FAILURE in 2h 03m 33s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/a1aefa9/DellEMC_PowerFlex_4.x_NVMeTCP/45 : SUCCESS in 1h 38m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"b4582822cd17c7919e68dcc0957b8adcc1141165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 05:54:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb9968786db04cd49f03fe12b5966706\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401bd7daf3b648d2b03da4d096846c81 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2e984eb62a64b45b2c3f9f0e702c908 : SUCCESS in 13m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f4f21f7da0b4a75aac011d876d49228 : SUCCESS in 10m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66592237c53741668e64417088bd0965 : FAILURE in 10m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b2e3baa2b5f4697ba41856e89206a06 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/f60785c1164d47e3ba76a636e080ad20 : SUCCESS in 40m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bba2d1f20dac4f7aa87ad235d66ef6d7 : SUCCESS in 1h 23m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bed530991cc340eda641b33cd857cf14 : SUCCESS in 59m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/931cf396978642be8dfccd940e5cd16a : SUCCESS in 10m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2a16b873ffa847dc97af423c0c235efa : SUCCESS in 16m 05s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fc071cfbc72c4feca8165f3e98eedf11 : SUCCESS in 14m 58s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/27f25c9e44204f3a9651dc8b3fdc9313 : SUCCESS in 4m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6241a6233d3845c3940cac72c121570f : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6d12209596b4e0c9043b41368b2d941 : SUCCESS in 5m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/54785c4114394b40bbd7a82226b00dec : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/468c3fc71cb847c291408c4df8450d25 : SUCCESS in 43m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50d17fb7cbff4f3cbec30b7e8b2be3ee : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/594daf1f45fe4daca2a951d77cee75ae : SUCCESS in 1h 49m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/534b4280ad634ee29f62cf5062e447d1 : FAILURE in 2h 11m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9de0a1cbe299464fb6d3b0e103777f6f : SUCCESS in 1h 46m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8104fc0cd80e458d95649c4fc44bde35 : SUCCESS in 1h 55m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9b820700908340b0a3d2f8106df443b7 : SUCCESS in 39m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/834123f2234f43c4965d0016b9161c78 : SUCCESS in 1h 08m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2191f7c4263c4cefb17ad309549ba175 : SUCCESS in 1h 24m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f3260257031b47538268a2da5fec6046 : SUCCESS in 28m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/30ca923eaad3403ea77aa61d14939f25 : SUCCESS in 58m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/216a9350f472457d92cf0586a6b4f9c9 : SUCCESS in 1h 15m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c0e43e4da1a4309b9f3b9a0b9e6753b : SUCCESS in 1h 38m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb005d201fc242a692125e5cb41d3bff : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":8},{"id":"bfa6fdd3090bcce81f048258ee97a5d88333bfeb","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-12 06:09:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b91cff520fcda10e4c74799a06afff5067ea9eed","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-12 06:36:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/8/upstream-check/cinder-tempest-plugin-ontap-fcp/80f6b0e/ : RETRY_LIMIT in 3m 46s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/9644f19/ : RETRY_LIMIT in 3m 36s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/8/upstream-check/cinder-tempest-plugin-ontap-nfs/131670b/ : RETRY_LIMIT in 3m 51s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/2e546bf/ : RETRY_LIMIT in 3m 26s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"bd695d7c92033f7baa851cf2c5805c9b3df0fbad","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-12 07:35:49.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerStore_iSCSI/3d81b8d/DellEMC_PowerStore_iSCSI/1231 : SUCCESS in 1h 15m 57s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerStore_FC/2af05fa/DellEMC_PowerStore_FC/1402 : SUCCESS in 1h 00m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerStore_NFS/7f4633c/DellEMC_PowerStore_NFS/921 : SUCCESS in 1h 03m 56s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerStore_NVMeTCP/5b83b12/DellEMC_PowerStore_NVMeTCP/782 : SUCCESS in 1h 41m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"01fd315662b13199ea18a08b3968c63fc04b3154","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-12 07:37:28.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_Unity_FC/678d81c/DellEMC_Unity_FC/1661 : SUCCESS in 1h 13m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_Unity_iSCSI/a139322/DellEMC_Unity_iSCSI/1692 : SUCCESS in 1h 42m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"574e46cddc03438623cc8e266b7a30c6456e60ac","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-12 08:14:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"da22c555b7e31db9bccf69e2c76ec1641d3c4bff","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-12 08:22:36.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_3.6/65b2e8d/DellEMC_PowerFlex_3.6/1169 : SUCCESS in 1h 43m 32s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.5/f55f54a/DellEMC_PowerFlex_4.5/140 : SUCCESS in 2h 12m 10s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/0df4c69/DellEMC_PowerFlex_4.x_NVMeTCP/47 : FAILURE in 1h 48m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"d4ca1a90e5d8a5e5a8185ed62e9cd242c625d7c2","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-12 10:11:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"32f1c6c57f2d3f895e84542bb9fca5d01350230f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-12 12:28:26.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_3.6/92f84d7/DellEMC_PowerFlex_3.6/1170 : SUCCESS in 1h 43m 06s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.5/b9dc219/DellEMC_PowerFlex_4.5/141 : FAILURE in 2h 01m 36s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/c3e3b1f/DellEMC_PowerFlex_4.x_NVMeTCP/48 : FAILURE in 2h 17m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"c4f3b396d2d90c3ba9a1a527e75eb2c3aff53d7e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-13 13:55:12.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/8/20240413080228/NEC-FC : SUCCESS in 9h 06m 24s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/8/20240413080228/NEC-ISCSI : SUCCESS in 14h 52m 40s","accounts_in_message":[],"_revision_number":8},{"id":"5730e5809a2f52618a895f95f14a051e80d73053","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-15 02:09:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6d3c045b5a6f669d1b1d64d97fcb4005597c1b7c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-15 04:26:33.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_3.6/c377887/DellEMC_PowerFlex_3.6/1177 : SUCCESS in 1h 35m 46s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.5/dfa8f0d/DellEMC_PowerFlex_4.5/148 : SUCCESS in 2h 05m 58s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/4faa808/DellEMC_PowerFlex_4.x_NVMeTCP/55 : FAILURE in 2h 16m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"442e49cec07acf92aad39dc6a59fc0bf1d3c21d8","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-15 06:55:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"868815b3a27c70b519b70e152ce4c07135421bb7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-15 09:15:07.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_3.6/e35bf3c/DellEMC_PowerFlex_3.6/1179 : SUCCESS in 1h 31m 45s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.5/bbd5061/DellEMC_PowerFlex_4.5/150 : FAILURE in 2h 06m 49s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/6c7e64b/DellEMC_PowerFlex_4.x_NVMeTCP/57 : FAILURE in 2h 19m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"a37f776b4c4a773427452df65f8bf53a74ca604f","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-04-16 02:10:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a7ed648b4400d96cb473ab3427148daa9d6f7a88","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-16 05:07:35.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_3.6/a2b1096/DellEMC_PowerFlex_3.6/1187 : SUCCESS in 1h 50m 08s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.5/f3d62d9/DellEMC_PowerFlex_4.5/158 : SUCCESS in 1h 53m 30s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/521d64e/DellEMC_PowerFlex_4.x_NVMeTCP/65 : SUCCESS in 1h 53m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"a25344d3ce452da8be3b9d38c775fc5a91944fc8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-04-17 17:11:17.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/912058-8/HPMSA-iSCSI/ : SUCCESS in 2:59:15\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/912058-8/HPMSA-FC/ : FAILURE in 37:13.71","accounts_in_message":[],"_revision_number":8},{"id":"0849166d004cad9fd8afdc76abdefd022112b986","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-04-21 10:24:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/912058-3/Lenovo-iSCSI/ : SUCCESS in 2:07:33\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/912058-3/Lenovo-FC/ : FAILURE in 3:33:52","accounts_in_message":[],"_revision_number":3},{"id":"98563ad2e67f5cac94d3922cd0b6ffbea2a65e53","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-05-09 02:01:17.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/912058-6/Lenovo-iSCSI/ : SUCCESS in 2:07:43\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/912058-6/Lenovo-FC/ : FAILURE in 37:48.46","accounts_in_message":[],"_revision_number":6},{"id":"ffcf940d43a7c10adfc5c9179d71466f37b33b4f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-05-11 16:15:08.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/912058-8/Lenovo-iSCSI/ : SUCCESS in 2:07:33\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/912058-8/Lenovo-FC/ : FAILURE in 3:34:32","accounts_in_message":[],"_revision_number":8},{"id":"d33e4ab56e5e56490c5c070109e41f604cf9406c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-28 18:10:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/8/20240728230621/NEC-V-ISCSI : SUCCESS in 3h 39m 46s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/912058/8/20240728230621/NEC-V-FC : SUCCESS in 4h 03m 57s","accounts_in_message":[],"_revision_number":8},{"id":"4c1ff54b4227152cfefd60019b236e0a8067f6fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-07 14:36:10.000000000","message":"Patch Set 8: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f1f0e59544a5c977335edff89371e2d2b8fc2f2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-19 09:52:11.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"1e61d89954a045866855fa4d44c3cb5d6a926058","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-19 09:52:24.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/317/","accounts_in_message":[],"_revision_number":9},{"id":"205a7552cce66cc847a9d6cf3d954f331e0da0f1","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-19 09:54:02.000000000","message":"Patch Set 9:\n\n(17 comments)","accounts_in_message":[],"_revision_number":9},{"id":"901999ef1bbe0fdb874022f4feef66b00ee1a745","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-19 09:55:15.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/317/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/9/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"8599fc6eea75a0be1a6b1eb2a2b8681f378925ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 10:17:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7926ead14cd84fdc89057d4ec25f8354\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/084b66f8d8da46069a27f3c7544d909d : SUCCESS in 23m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/41c03aa5594d41dabd386ce9805a09de : SUCCESS in 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7c9a449aa443ed827b6413375f9e01d5ce1298c0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-08-19 10:23:15.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/912058/9/ : RETRY_LIMIT in 8s","accounts_in_message":[],"_revision_number":9},{"id":"9a264f4108f99473a1a33207b117a6e066145b0b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-19 10:24:46.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058 : FAILURE in 31m 09s","accounts_in_message":[],"_revision_number":9},{"id":"293ce0b82b89be8ca6f65a9b5774906112c55bc2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-19 10:26:31.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 33m 00s","accounts_in_message":[],"_revision_number":9},{"id":"f89b2cee69e7ca6a336624142dbe69401995c7aa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-19 10:33:05.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : FAILURE in 31m 46s","accounts_in_message":[],"_revision_number":9},{"id":"2253e9db720a6a3d6b0dbb02f3d8e2c5e3d68e4d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-19 10:44:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/329/.","accounts_in_message":[],"_revision_number":9},{"id":"414bab0ae30543dcc3afafab37b224ada21378fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-19 11:47:55.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cc77e8ee2a444689a20d06697c740850\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6017ec6be04b4b0db76d1c0aa3a3e451 : FAILURE in 38m 24s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1a35ebfdb8e84e27861967401b5b1f77 : FAILURE in 1h 37m 12s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6aebc7b9c52a4134b2fa425cff82bcb7 : RETRY_LIMIT Ansible setup timeout in 1m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/32c05d89786246149ab23235dd2d8231 : FAILURE in 1h 48m 40s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher... (truncated)","accounts_in_message":[],"_revision_number":9},{"id":"c425458e709ca97bbeafd3a7f7e845dd92a40bd7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-19 12:01:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1db770696e384f669cd46e54156ce01e : SUCCESS in 1h 12m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5b5105bc35134074898ea600d801d4c4 : SUCCESS in 1h 14m 21s","accounts_in_message":[],"_revision_number":9},{"id":"6403017acb22688e8a673f8a67754e31e44914aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 12:39:19.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d4fcaeb6d76486db8ba1bcfc201bc5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed3845a0ee44a80995030395c8ed4a9 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82b6811536d44732be962bb2678f6fef : SUCCESS in 9m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc5029f2b555416f9c403bdeb8a2818d : SUCCESS in 8m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a956c113b6c144f2863fcf617e5ef091 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78686da019fc4022a5d0cb99cbbc915e : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/4ceb7c7387b646da818e1fb897c605e3 : SUCCESS in 1h 09m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92d1e7087a0c406bb2b7f189796c5258 : SUCCESS in 1h 28m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ee5a4f9f5a94692baeea260f583f35b : SUCCESS in 1h 01m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/300143f8fa8044188172afefce8caeb6 : SUCCESS in 19m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3c7a56fee54247f295168dd520a769bc : FAILURE in 16m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a2b5d90a1c234b97bce0d9a5c3dc7478 : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ff227b8982e4f7e8de8c57737b6664d : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ffa53afae3a14179b45fd51723f29d79 : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/081e99ca2d8543fb8dc236521067cbe8 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b71407e8d2264791a5da722d01cb2932 : FAILURE in 36m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/291ca3aeb5034978ba7387a3344df65c : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7fa3d8c08cb743e5aa14135b9b7d1177 : SUCCESS in 1h 52m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/32c93db1dfbf44f5baf8f7bb182dc174 : FAILURE in 2h 34m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52e095a98de2425b89f0e7512dbc0854 : FAILURE in 2h 19m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9a176e29df5d4b11aca750effd470042 : SUCCESS in 2h 15m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/05ba64df54f04d338e7eb3f36d2f6838 : SUCCESS in 35m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1b7a274f3b3c414ebca810b958a25c9a : SUCCESS in 1h 15m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0bf71d0a72e94161a491b4b258f6127b : SUCCESS in 1h 53m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0280a8944b9341d4a1dd20745a959778 : SUCCESS in 50m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4c9f1ba7c6d14005bf197e1b6193ba63 : SUCCESS in 1h 44m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b5d4f53e9b3d44a697160bf094fe0a7d : SUCCESS in 1h 40m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/12a5fac49b4d462d8d1c4d0eda60e388 : SUCCESS in 1h 30m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/907be8f21dbf4d38959e7422bbf3f335 : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":9},{"id":"e0ce057745695c59e68f86a3e827cd1a81feaa60","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-19 13:05:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8d58e0d39af9f837c7f59d2a6ba7493866774703","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 13:55:01.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/811c416b15354b7c8487ce49907fe5cf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73934c8d46ab4fc08e42e78323cf2aaa : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/edc063dbe0644d86a3079ab47b29dceb : SUCCESS in 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5c92e073ff097935db8b1cae3ae2ce47eba5627f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-19 14:06:53.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/915/.","accounts_in_message":[],"_revision_number":9},{"id":"cdc5b5be9b674f0bfa74dfe90ff41799b4375352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 15:29:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cca4ba3b71584c559b0287e3465ede12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373fbaba8cb245cca130045cf9ebd9bb : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20822a36a71942ac80e3124ce3c35e79 : SUCCESS in 12m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/209c7c604e424abcae8a08da18626efc : SUCCESS in 10m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e6a3f839cb54418bc3712e15d7f7172 : SUCCESS in 12m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab575ea4f2a4a1eb8e292d8411818ab : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/56b6b35c45274c568aca344f8eef5cb4 : SUCCESS in 52m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f4efee7e0884a07b56196110f5ab649 : SUCCESS in 1h 31m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/228ae74ebed542539055b7c5ecc2267e : SUCCESS in 57m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/720f4df84cfa4f488278e69e7d00012c : SUCCESS in 22m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b52c2612761246989b442f27eca70522 : FAILURE in 15m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/88aa5e09f5044d7c9b334d90c1d8dd06 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8817f0347b5e403fbd1dbe65c804d9d5 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47ab3d8fe2434c1da4f1fb709b5b3972 : SUCCESS in 5m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/012ef8fcf3b14261b3aea7ea24a07c58 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cfd1caec8f83479cbfbcfbfbe91b2ca6 : FAILURE in 32m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06fd9a3f44f948fd852d9e2aec05e2f2 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c42d8eed54014a3d87d4f49e4bbc29a5 : SUCCESS in 1h 50m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1296f1a5c2684754a70d044c0fbf5876 : FAILURE in 2h 23m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1ceff0ce83d4ea0bfefaf57e70aeccb : SUCCESS in 2h 10m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/603dc6cff844407f8fb1c8f0c02a584f : FAILURE in 1h 58m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/63143e4e570440239f15ebfc7bd9c68b : SUCCESS in 34m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2063f703846b49e6ae33da4398dd6211 : SUCCESS in 1h 10m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d975b38c03fc48648390f26c6cc1608e : SUCCESS in 1h 26m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/409ae9ddca1444b2af90c48313cc13fb : SUCCESS in 31m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/326c6d788e8a497d932902d7a515052a : SUCCESS in 1h 16m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e161e473f2544e88ae72228976640986 : SUCCESS in 1h 41m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cb66fd3a10f54b4d83ed7cf00efa6db9 : SUCCESS in 1h 27m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85b656e9e5b94a4b801ad91e8c34d8e5 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":9},{"id":"f2e344d5b01673b5957b2a840ceb5711725d7478","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-19 15:35:22.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_08_2024_15_29_49\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"085485c31e7501b2527545d9fd5053b3e8fcea76","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-19 15:39:36.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/addf0f0/ : RETRY_LIMIT in 4m 27s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/9/upstream-check/cinder-tempest-plugin-ontap-nfs/ab6bd77/ : RETRY_LIMIT in 8m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"e3eb9f3d7b842292c0e3630507deac54faa3e676","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-19 16:00:08.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/9/fc : SUCCESS in 0h 46m 21s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/9/iscsi : SUCCESS in 0h 56m 52s","accounts_in_message":[],"_revision_number":9},{"id":"54b384dff00898b0ded93c01bdf01efb5cec4dcb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-19 16:17:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aab15474b3c6443494f78c71ef49c488 : SUCCESS in 1h 12m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3ad0625c840c4e0ea2c6653dc52198f0 : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":9},{"id":"f52cd0ed79586ff2630344a4d05954819241cbde","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-19 17:09:52.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_Unity_FC/def2ffe/DellEMC_Unity_FC/2446 : SUCCESS in 2h 01m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_Unity_iSCSI/b0173b7/DellEMC_Unity_iSCSI/2449 : SUCCESS in 1h 53m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"58f22bcff96e45f98be5858c50da799eedabf33e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-19 17:40:56.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_PowerStore_iSCSI/1919f50/DellEMC_PowerStore_iSCSI/1615 : SUCCESS in 1h 24m 58s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_PowerStore_FC/6c7d015/DellEMC_PowerStore_FC/1811 : SUCCESS in 1h 03m 13s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_PowerStore_NFS/c288a9c/DellEMC_PowerStore_NFS/1307 : FAILURE in 1h 13m 25s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_PowerStore_NVMeTCP/a3c955e/DellEMC_PowerStore_NVMeTCP/1167 : SUCCESS in 2h 10m 57s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_PowerStore_NVMeFC/b202a14/DellEMC_PowerStore_NVMeFC/131 : SUCCESS in 1h 10m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"1300a9542de53f502b0344f86711d2d8cc83ebbb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-19 19:25:13.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_PowerFlex_3.6/dd6d660/DellEMC_PowerFlex_3.6/1923 : SUCCESS in 1h 53m 38s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_PowerFlex_4.x/3294b67/DellEMC_PowerFlex_4.x/804 : SUCCESS in 1h 57m 23s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/abece8b/DellEMC_PowerFlex_4.x_NVMeTCP/775 : FAILURE in 13m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"49f22f32ef540d0d220d90c0c9490155d123c1cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-20 03:45:12.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"db6a411c0edfbf1423499467f9be1bcc81e05e45","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-20 03:45:21.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/326/","accounts_in_message":[],"_revision_number":10},{"id":"4508a08663ab3a250e3c8b288ab58191fa8ad1dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-20 03:45:23.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/326/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/10/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"c612baa0ce59647fae4da5d0888161803fb6f8f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 04:20:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/537a151911bb40419139e4b8c362846d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9ae3ecfe5754d859d1e5d549cd0fb32 : SUCCESS in 33m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cae14969aeef4c4c8daf8daffab00f0e : FAILURE in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"69ae20cef590dd56ca9fcc9b7d78ddd0e06c9ec9","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-20 04:30:49.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058 : FAILURE in 34m 46s","accounts_in_message":[],"_revision_number":10},{"id":"1cb694974a37f4130edb2d93e1cc36af3f692750","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-20 04:36:52.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 34m 36s","accounts_in_message":[],"_revision_number":10},{"id":"306d1ec8c969cc92391d233827276a5a0d50347e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-20 04:49:20.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : FAILURE in 32m 40s","accounts_in_message":[],"_revision_number":10},{"id":"0e02f066fc38151e8fe6d7eed64d3f2e7fd33bd9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-20 05:02:11.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4e9a05185b2940d597f69ced195851b9 : SUCCESS in 1h 14m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4345ac262ede4af1a519100618223d35 : SUCCESS in 1h 11m 35s","accounts_in_message":[],"_revision_number":10},{"id":"b9fbe3e7d66c66e813f56e275e24601dbdbf3c0a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-20 05:34:09.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c83fe70625c041a68679f69db0741e32\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd8cf460e48a45ca8bc04215bdb5a46a : FAILURE in 42m 11s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0195ceb8de084996b4bacad9afa383e9 : FAILURE in 1h 44m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/79abff30b1d84485b69d8f4e09f0dadf : RETRY_LIMIT Ansible setup timeout in 1m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/372b4eac31904078907ea448db7eb3e5 : FAILURE in 1h 21m 15s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"cd8adea0f738008f3882f0d42a24b812a8211c72","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 06:12:16.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/924/.","accounts_in_message":[],"_revision_number":10},{"id":"1b3cfb0daab41c66ee44f7201b8eabe7fb3da092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 06:19:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7befd94673fe40d9ad0d2d5e730120b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6acc3f101c407c88e0578a15dd4524 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffdcb110a38b478487f93cd47b270a50 : SUCCESS in 11m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdeea44ad03f466ba6f2d9a88cfcf1b7 : SUCCESS in 11m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3080e3662c4541aca9c3118e5d69bde8 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409af40ecb8e4d1abcd0bc11ec0af3e4 : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/b538afb97b9d43c49984ad6a2b34c68b : SUCCESS in 56m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d226a7dab1a541e590792ead3bdf416f : SUCCESS in 1h 24m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f03bf06f83a14997b6097548edb2c15f : SUCCESS in 54m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d906a4159664fc0a8c85e211f196fdf : SUCCESS in 19m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1340c1483acc4d3ba24e9ad9f4e3539d : SUCCESS in 15m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aa148119331c4d66a63d6aa133f50e40 : SUCCESS in 4m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/028f30a3212f41ffaa4025f27c802c2c : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea594b20631b4d25871df5ffe89e642f : SUCCESS in 5m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/adde6c24c0704b2db624fadc84899670 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8bd0e37d8c13475ba943ca47fe4b0287 : FAILURE in 35m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29aeaa0a08104e3f9e4c4ae0d72b06d6 : SUCCESS in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4885439c7f0b4871970d7f93154516ed : SUCCESS in 1h 59m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/17b146aed8794b91834aeaf50ea89453 : FAILURE in 2h 24m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f655cca8e374736b2a1c1c3bbd90e6a : SUCCESS in 2h 02m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/48e40a97d524445daf0a3c90f2e55383 : SUCCESS in 1h 56m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9ff48bed219d4e92ba33795972f6acbf : SUCCESS in 38m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60ab89c2d6ce4020b2eab5c796c3cef8 : SUCCESS in 1h 24m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/179fdca223af496b92c1464c9b124cbe : SUCCESS in 1h 11m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4590e1781f4642ef822cb2e187576c8a : SUCCESS in 42m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b52313b4bd554fa79080c096ab6be878 : SUCCESS in 1h 04m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b3440ce484114ab3bdbb8040d6b9713f : SUCCESS in 1h 29m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c66757a86484625a97dc2bde97ceab0 : SUCCESS in 1h 16m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c446cd4972474118b77bcdfdb7ae9bed : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":10},{"id":"ee86520f80fd52929e111a4e2d70c74a7de32543","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-08-20 06:25:30.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/912058/10/ : POST_FAILURE in 1h 36m 15s","accounts_in_message":[],"_revision_number":10},{"id":"0ef815123ce812c1f640ebc1ee1bdca0ae29031e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-20 06:25:34.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_08_2024_06_19_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"e557e77184db8555fe989ed2ca486a877821fb7f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 06:39:07.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/337/.","accounts_in_message":[],"_revision_number":10},{"id":"7da1c2a0704c00bd27794a60498bd8683ad591c5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-20 06:39:30.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/d559a50/ : RETRY_LIMIT in 3m 56s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/10/upstream-check/cinder-tempest-plugin-ontap-nfs/f59c069/ : RETRY_LIMIT in 3m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"841c5f37a9164d3db14f590654c88e6ca86ec76b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-20 07:19:49.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_Unity_FC/f0dd8a7/DellEMC_Unity_FC/2454 : SUCCESS in 1h 32m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_Unity_iSCSI/c46e6db/DellEMC_Unity_iSCSI/2457 : SUCCESS in 1h 48m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"5676218ccf1ccc7a12b5376053641b488a524c70","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-20 07:42:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"63dbe59e1f7884a3c4c94f0e2301af7f4bac42c2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-20 08:10:26.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerStore_iSCSI/18936be/DellEMC_PowerStore_iSCSI/1619 : SUCCESS in 1h 36m 47s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerStore_FC/795150d/DellEMC_PowerStore_FC/1815 : SUCCESS in 1h 00m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerStore_NFS/93c3a7c/DellEMC_PowerStore_NFS/1311 : FAILURE in 1h 21m 36s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerStore_NVMeTCP/5d5c3ca/DellEMC_PowerStore_NVMeTCP/1171 : SUCCESS in 1h 50m 40s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerStore_NVMeFC/15abcba/DellEMC_PowerStore_NVMeFC/135 : SUCCESS in 1h 16m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"990d298b6f43ee1ead47b7137758037f79bdde8a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 10:17:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/339/.","accounts_in_message":[],"_revision_number":10},{"id":"b7202d2ae80f6e5d966c8a4b5034c78157706044","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-20 10:39:18.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_3.6/d6f40e9/DellEMC_PowerFlex_3.6/1929 : FAILURE in 1h 48m 46s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_4.x/be76218/DellEMC_PowerFlex_4.x/810 : SUCCESS in 2h 11m 24s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/4ae866f/DellEMC_PowerFlex_4.x_NVMeTCP/785 : FAILURE in 2h 56m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"86c42621b21375f0a8a6b4281fcbcb011813c6e2","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-20 13:46:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"561a2d98b7fb9eb2e53bf6a6ab406c22b5363437","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-20 19:24:37.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_3.6/a23c862/DellEMC_PowerFlex_3.6/1932 : SUCCESS in 2h 02m 40s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_4.x/c27f388/DellEMC_PowerFlex_4.x/813 : SUCCESS in 1h 57m 48s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/02ce26f/DellEMC_PowerFlex_4.x_NVMeTCP/788 : FAILURE in 2h 55m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"e118bdfe6190e74d5fd5555e1f8a7bdb5e561759","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-08-20 22:24:53.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerMAX_FC/80dd0ef/DellEMC_PowerMAX_FC/424 : SUCCESS in 6h 25m 41s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerMAX_iSCSI/4407b49/DellEMC_PowerMAX_iSCSI/446 : FAILURE in 7h 41m 29s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":10},{"id":"61ed9a18677ef82c75190df11f75f27d0f9f9a53","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-21 02:27:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"99e081d566fd6b8cc8bfc0b2653e9eed84250b7b","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-21 03:18:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6b943f70ab0e3f56be6ba24b02aa348ae0911641","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-21 06:13:41.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_3.6/2db2bf7/DellEMC_PowerFlex_3.6/1940 : FAILURE in 2h 04m 54s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_4.x/5874ecf/DellEMC_PowerFlex_4.x/820 : SUCCESS in 2h 16m 21s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/18ab650/DellEMC_PowerFlex_4.x_NVMeTCP/795 : FAILURE in 2h 55m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"ce826dab4a602e807db7b185aaf92f9f00d0be0b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-21 08:11:41.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/10/20240821132126/NEC-V-FC : SUCCESS in 3h 42m 26s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/10/20240821132126/NEC-V-ISCSI : SUCCESS in 3h 50m 11s","accounts_in_message":[],"_revision_number":10},{"id":"5428eedd1fc919b5c0f760c6afa8d4634bff4b1c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 02:34:37.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/10/20240822020342/NEC-ISCSI : SUCCESS in 9h 07m 47s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/10/20240822020342/NEC-FC : FAILURE in 9h 30m 52s","accounts_in_message":[],"_revision_number":10},{"id":"563b5099c007511e9fb9d78f0842c2bd3f3c4d1e","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-22 09:44:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0e5b14bf97acda752e51c3b996c12b2dfa1c4b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-22 12:47:02.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"0b28cde02155e7e99ce54b4b34162e628c6e7560","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-22 12:47:13.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/370/","accounts_in_message":[],"_revision_number":11},{"id":"4003ebd6a165497b635b61372f6605591c19b4dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-22 12:50:35.000000000","message":"Patch Set 11:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/11/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"98824cbec09f53a30c925f3d3075c213c3f7b271","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-22 13:02:56.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 15m 01s","accounts_in_message":[],"_revision_number":11},{"id":"7d40a245bd33f9e68d683c2dcff3f5ff12353e62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-22 13:10:31.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e819904a77cafb93faa6ebe64642fb49f11ea3d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-22 13:10:43.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/371/","accounts_in_message":[],"_revision_number":12},{"id":"53c53de070286befa51e499e51e2c30908d048b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-22 13:14:43.000000000","message":"Patch Set 12:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/12/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"8d5d5219774fb199dd299cb0c3df872f8874d960","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-22 13:26:21.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 15m 00s","accounts_in_message":[],"_revision_number":12},{"id":"79681a35f3caf2e4f8dc019f4b9a0a3082ca4566","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-22 13:56:03.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058 : SUCCESS in 43m 40s","accounts_in_message":[],"_revision_number":12},{"id":"d9f9b1c43fe8607ebea5d8055444c2d3b30d1b78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 14:05:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/378/.","accounts_in_message":[],"_revision_number":12},{"id":"55f4346a2d0dd2cef5aa8924e08be330d6ac6677","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-22 14:14:09.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":12},{"id":"cc78fc8abc3727957064a3810065cf8b3e83e0b7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-22 14:36:05.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f6044cfd40ee48c1ac4576a7478592ac\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/38dff4c837674a7691c6c7175a813c05 : SUCCESS in 39m 10s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1b8d7628d9204d16a97cbd8de22f5584 : FAILURE in 38m 48s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7f45f8cc508c46b3abd09ec5719bf2cf : FAILURE in 1h 25m 02s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0ffd8651a4c049b592f0a5d49344b59e : FAILURE in 46m 00s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"7cf3d2c36308ea5184cbd417ac7c4c0fef9d78ee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-22 14:38:26.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/401e794e54604160b4c235b162af6896 : SUCCESS in 1h 25m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/190a1dab42914d838223e4e055a3a0d2 : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":12},{"id":"8cb80821984f408e4e7db5677671ec09ff85a767","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 14:44:45.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/966/.","accounts_in_message":[],"_revision_number":12},{"id":"9be705dc22efd65aac83a1255d60d4d1b00e0a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 16:03:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4f75f1e620149f5a52642a2e25e6566\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16223be5fc24ee9ae5bd044e181917d : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dde2d4c32ed478f9c7eecf4cd8c0674 : SUCCESS in 10m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdbba7b682254883bed20bc2a474c929 : SUCCESS in 10m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/946796c1eedc4d9281c94840a613ddb8 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a97ee2892c43fcab0aa20d745fca07 : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/5501ab268a344a3588ca8ebc9e6367a3 : SUCCESS in 1h 11m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/adb619d2607e4fe6b40456bd16fcf19d : SUCCESS in 1h 27m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ceea0227c71b4c0badd8a029831c9ccc : SUCCESS in 55m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2785edafb53444cabccbbea233518811 : SUCCESS in 17m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1125655119a24a45bf4b2fe988b300fe : SUCCESS in 17m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7cd3d0d4c7bd4a179f4e83d34298e5e3 : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8161dc50578f458591f7590b229c2a1f : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad431fe5352f46c5b18903b0600b970c : SUCCESS in 8m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0b22ad0d9980461d84cde4f820dac7fe : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f12206e1200c495693bf8fecd3e62f9b : FAILURE in 35m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/18b1d54e99bf4c1d843bf0f8fa7fb04c : SUCCESS in 8m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea41ea54dfce456080ecdc095b7791b6 : SUCCESS in 1h 53m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/05bb8087dd12433c866be965f049a220 : FAILURE in 2h 32m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f3ac9b620294d859b53d65bc1f41cad : SUCCESS in 2h 46m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5d406196203e4bfd96d353883a4a66f9 : SUCCESS in 1h 50m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1fe086e53bcf4328978cbf641fccafb3 : SUCCESS in 32m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60bc812d6d8a4128a106641dd2f7ef24 : SUCCESS in 1h 01m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b27c764185134081a6f750b682196974 : SUCCESS in 2h 24m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8de8bcc64be74b0b9120cbc3dd381135 : SUCCESS in 42m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4bcc69ab9ec94bbca689c3900616a9dd : SUCCESS in 1h 14m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c2779ae145884f32981e5e545639fde5 : SUCCESS in 1h 26m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aae58d69a82940439c76546094c2f779 : SUCCESS in 1h 33m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e052aa5a56346a1a3f7924d951c30fa : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":12},{"id":"678fb89b126a359c1011d1a5bfb3b037df66b59f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-22 16:09:21.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_08_2024_16_03_46\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"7de3eca463332ccc3e55374993201facf1c30178","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-22 16:23:28.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/6a6f3ba/ : RETRY_LIMIT in 3m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/12/upstream-check/cinder-tempest-plugin-ontap-nfs/22cdcf0/ : RETRY_LIMIT in 3m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"e6cc868b3d83840992613d855c14313b3c170b7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 16:54:38.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5a0245bcc074a45aea0cad3eeaa0a5f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62c554f654734f3f9d189728575c6e63 : SUCCESS in 21m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21224509f60442e1ac3631a4e5f5090d : SUCCESS in 19m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a5d6e390f12bff763ee80c3210b2b2c7c3ce9690","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-22 17:04:35.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/bca3237/ : RETRY_LIMIT in 7m 25s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/12/upstream-check/cinder-tempest-plugin-ontap-nfs/d78fee0/ : RETRY_LIMIT in 3m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"a7e5e14b1edcc52dd5ed094f214f759f31096de3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 18:40:57.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/3eca773/DellEMC_PowerFlex_3.6/1945 : SUCCESS in 2h 22m 37s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/2df9b05/DellEMC_PowerFlex_4.x/825 : SUCCESS in 2h 00m 50s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/36317bc/DellEMC_PowerFlex_4.x_NVMeTCP/800 : FAILURE in 2h 53m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"4f2e1a5f09def6375775753639f44ded5aaa47d7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-22 20:26:15.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_iSCSI/2b3ca9f/DellEMC_PowerStore_iSCSI/1631 : SUCCESS in 1h 25m 47s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_FC/dd1ae45/DellEMC_PowerStore_FC/1827 : SUCCESS in 1h 09m 00s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_NFS/6daa286/DellEMC_PowerStore_NFS/1324 : SUCCESS in 1h 18m 38s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_NVMeTCP/466ec65/DellEMC_PowerStore_NVMeTCP/1183 : SUCCESS in 2h 05m 45s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_NVMeFC/97e802c/DellEMC_PowerStore_NVMeFC/148 : SUCCESS in 1h 05m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"cb08f40f4aae68951390015641857a78cd76b477","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-22 22:22:22.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_Unity_FC/6ebcdf6/DellEMC_Unity_FC/2479 : SUCCESS in 2h 22m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_Unity_iSCSI/27aace3/DellEMC_Unity_iSCSI/2483 : SUCCESS in 2h 21m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"a14f1870f35aa3279b3375f8f203e6fc0a9cfbaa","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-23 05:03:45.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e4f57f3f67d44af48606f03fc717c2cee66e54d4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-23 05:04:36.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/b407345/DellEMC_PowerFlex_4.x_NVMeTCP/807 : ABORTED in 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"6c77c58e76105af847a6e3d9a2f5ab3abcdc2442","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-23 05:04:37.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a5a9ff5cfb8b8dc9d691a294fbe09299eed3a8bd","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-23 07:59:03.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8995a9c0ee40e7eb346190f069b42d00227c7a02","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-23 08:10:45.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"df62ae9897b0e41d5b7335888541c421c04c0d2f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-23 14:06:08.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/e9f06d2/DellEMC_PowerFlex_3.6/1951 : SUCCESS in 2h 30m 05s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/a552fcc/DellEMC_PowerFlex_4.x/830 : FAILURE in 4h 06m 49s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/4566f12/DellEMC_PowerFlex_4.x_NVMeTCP/809 : FAILURE in 5h 55m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"8e32322d2754f6f9c29da4bec98787885a98fc4a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-23 15:00:08.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/12/fc : SUCCESS in 0h 44m 38s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/12/iscsi : SUCCESS in 0h 58m 40s","accounts_in_message":[],"_revision_number":12},{"id":"fadf616d88db204c7e0091d1e11135d3967d9dae","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-24 16:46:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"630e081c96e632b65073fa8c231e46f0a756416a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-08-24 18:23:37.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerMAX_FC/5fb8eb9/DellEMC_PowerMAX_FC/436 : ABORTED in 10h 30m 01s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerMAX_iSCSI/1db54ac/DellEMC_PowerMAX_iSCSI/458 : ABORTED in 10h 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":12},{"id":"8e37508b9cbfe69d47a60a443bd333ad9e9895d4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-24 19:29:21.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/48cf6f9/DellEMC_PowerFlex_3.6/1956 : SUCCESS in 2h 13m 54s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/6c92a60/DellEMC_PowerFlex_4.x/835 : SUCCESS in 1h 55m 37s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/3cda6e1/DellEMC_PowerFlex_4.x_NVMeTCP/815 : FAILURE in 2h 43m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"6ba07207ff4af663b1ab24209d7c32a4242dc800","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-24 22:29:51.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/12/20240824221907/NEC-V-ISCSI : FAILURE in 8h 07m 34s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/12/20240824221907/NEC-V-FC : FAILURE in 9h 10m 40s","accounts_in_message":[],"_revision_number":12},{"id":"5448d1cc5eb3d20d37fa2604278ddcec947191e6","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-25 07:14:14.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e1b230124b67d9894bddb0a201439657c0173d56","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-25 10:18:58.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/6c80a10/DellEMC_PowerFlex_3.6/1957 : SUCCESS in 2h 27m 51s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/092f2fd/DellEMC_PowerFlex_4.x/836 : SUCCESS in 1h 51m 36s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/78781a0/DellEMC_PowerFlex_4.x_NVMeTCP/816 : FAILURE in 3h 04m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"8e8f28575f952602ce2ad45ac67c1c2174040153","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-25 10:22:28.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4124c9d3ce349e6724a585771433cdb793775d16","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-25 13:04:08.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/337db1a/DellEMC_PowerFlex_3.6/1958 : FAILURE in 2h 11m 37s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/d09821f/DellEMC_PowerFlex_4.x/837 : SUCCESS in 1h 53m 53s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/65555ed/DellEMC_PowerFlex_4.x_NVMeTCP/817 : FAILURE in 2h 41m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"a58dfbfba6c364c61cdeb864ac2700f935bef84b","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-26 04:40:44.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"945316a9f9a80d906e8d4f92f830c4b47d1182ed","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-26 07:13:12.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/4fe5faf/DellEMC_PowerFlex_3.6/1960 : SUCCESS in 2h 08m 45s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/7a6f4bc/DellEMC_PowerFlex_4.x/840 : SUCCESS in 2h 19m 11s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/3f644cd/DellEMC_PowerFlex_4.x_NVMeTCP/819 : FAILURE in 2h 32m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"b23f73a243a773752742931bff068fe1f8ef9482","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-26 09:12:46.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0b44ffed7ba101bb0a1cadd7574eca706a8e7cbc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-26 12:36:58.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/6929cfc/DellEMC_PowerFlex_3.6/1963 : SUCCESS in 2h 02m 34s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/e27eaa3/DellEMC_PowerFlex_4.x/843 : SUCCESS in 1h 47m 18s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/2e4fda7/DellEMC_PowerFlex_4.x_NVMeTCP/822 : FAILURE in 3h 23m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"ece05e26dde544f7747d07de6c2954d9fc4afbf2","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-26 12:48:30.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b26418872c62d112b5e2e4dee9a999b652fbad86","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-27 06:23:52.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"844062f760fcc1c20b1232b6caf5d2b8608f2daf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-27 07:02:14.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/223d9e7/DellEMC_PowerFlex_3.6/1967 : SUCCESS in 2h 22m 25s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/7c754b8/DellEMC_PowerFlex_4.x/847 : SUCCESS in 1h 52m 17s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/18c5212/DellEMC_PowerFlex_4.x_NVMeTCP/826 : SUCCESS in 2h 38m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"d8f0f010a3a8f8abbc431a4d7afb8f34813b5620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 08:55:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c740939c7ed404aa0a3d18e77ab726d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19573626733842dfb7a0077c36eb1c4f : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36fdbc7d78854c2186ac3fe19dfcfb49 : SUCCESS in 13m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2a45d3dc93c4c94b468ac928b31aceb : SUCCESS in 13m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c9fb0b4068045d6bb15c650773fbc98 : SUCCESS in 14m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889c5ec2eec6482689db57aa7fa98030 : SUCCESS in 14m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/ce43dc8732a34e97a470dd42ba0a8d5e : SUCCESS in 59m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56e0c25963bf4fa592dd2fec5d825a6f : SUCCESS in 1h 32m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/971a4901af5c4ad8b9f5960669ff7e9b : SUCCESS in 57m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3afdde4c7104462aa59473845a93d23 : SUCCESS in 20m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/623799ce23ef4f7d975f1a8958788ce4 : SUCCESS in 18m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/15eca846a454419399b6abd706a7c631 : SUCCESS in 4m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8307b4fd9f4b4cc8a28b0c5ba2f5e76d : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/426326b422614d4c8b0bca7ab4defed9 : SUCCESS in 7m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/102e5d6bbfe24523a99d436ac6fae3f9 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c5ac142302cc4cffb7a63b646b66cd39 : FAILURE in 34m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7846d4320f8a4b53bba02458de52c8cd : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/638b2e28942b494cabe601241d2fa183 : SUCCESS in 1h 39m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ad494f904e14b1593b8299deebf6b29 : FAILURE in 2h 06m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c33c2ba0be3440ce99cb23fad44aeea4 : SUCCESS in 1h 40m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fcbf5cc2a57e4005ad8323a0035b19a6 : SUCCESS in 2h 07m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/384bbcefd569441792c9d96a4e92bf9c : SUCCESS in 31m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c851074547fc42abbf3779b5478055c1 : SUCCESS in 55m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4d353d5189b94009bf9b50745c222d6a : SUCCESS in 1h 32m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fa40848e7d4547269ab82a216dac011b : SUCCESS in 31m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5d05562d39d642cea46b802602e0c309 : SUCCESS in 1h 46m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/246ea41ea30446429a779431b8d08258 : SUCCESS in 1h 28m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90871c5c89444cde9056a707c6dd18a8 : SUCCESS in 1h 23m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b2780f752914fe9ae82713f8de8f120 : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":12},{"id":"f1e6fb27cc5c5583e2104208bbc0ba9d27ba0db8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 09:03:19.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5308d085414af4917bf81b0a232524\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3def9f82778a4655b737206fa622c645 : FAILURE in 21m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd56507899f24a1595c913b7596fb6d2 : SUCCESS in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6bb4d6d4d7d2e657af852295f9a6f3c53e6584d7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-27 09:52:42.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5d193a1af80645fe81f23fb4d04c6266 : SUCCESS in 1h 18m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5710fc7fb29645c9b2b36529fb5f0b2e : SUCCESS in 1h 17m 09s","accounts_in_message":[],"_revision_number":12},{"id":"02b44988bf38825b109239c72a2c989a2865c96f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-27 10:07:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/c65888c/ : RETRY_LIMIT in 4m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/12/upstream-check/cinder-tempest-plugin-ontap-nfs/f0a4e92/ : SUCCESS in 54m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"6cd6f824f48d50cf10171586fa20fea83fd9cede","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-27 11:14:08.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_iSCSI/cc54dbe/DellEMC_PowerStore_iSCSI/1652 : SUCCESS in 1h 40m 47s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_FC/7e6d478/DellEMC_PowerStore_FC/1848 : SUCCESS in 1h 03m 10s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_NFS/104cc6f/DellEMC_PowerStore_NFS/1345 : SUCCESS in 1h 23m 05s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_NVMeTCP/0e51f19/DellEMC_PowerStore_NVMeTCP/1205 : FAILURE in 2h 18m 05s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerStore_NVMeFC/f639913/DellEMC_PowerStore_NVMeFC/169 : SUCCESS in 1h 14m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"3b32a160ad4e73aeab722956fcf2a94c8c7c44eb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-27 13:17:10.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_Unity_FC/734621a/DellEMC_Unity_FC/2509 : SUCCESS in 2h 06m 05s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_Unity_iSCSI/2d20654/DellEMC_Unity_iSCSI/2513 : SUCCESS in 1h 59m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"b5a500d8ca3652152649a792f7d13b6a2ee8cca5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-27 15:08:05.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_3.6/94e8cec/DellEMC_PowerFlex_3.6/1972 : SUCCESS in 2h 09m 24s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x/fed149e/DellEMC_PowerFlex_4.x/852 : SUCCESS in 1h 58m 28s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/04ce6ad/DellEMC_PowerFlex_4.x_NVMeTCP/828 : SUCCESS in 2h 12m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"5805d079b9a078f358c05725ec24e9885bcc1654","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-08-30 19:05:38.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"398f1524a1138e2048e415eba0d90d3bec1c6980","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-30 22:39:49.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a7f39c8171e6f7840cfa082913c5f9614ba5cd07","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-31 01:29:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e0cb7a8835eb7e4e5eaaf0375097f79f30d2b9df","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-31 03:12:57.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"12b24a50c68637be4ce410283105992463ed85d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-31 03:59:03.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"1c30715f621a34ac70f0b4ac4b3e1586617ed53e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-31 03:59:11.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/447/","accounts_in_message":[],"_revision_number":13},{"id":"8ce4b2e466259c73a4e3ccdffa4bed50838a6a94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-31 04:03:46.000000000","message":"Patch Set 13:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/13/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"8b2961ddbd8782c8ad81b68905a1b9bd4b206cda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-31 04:20:22.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0581246c9cc40eebf20914b5ccafb65\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1d6f40a4cde46b8988cefc04c327f7c : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d2b28965d1814601b3e2e28df994e30c : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5456767b9460398a6106d0862d9729b8449c4485","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-31 04:40:49.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : FAILURE in 40m 09s","accounts_in_message":[],"_revision_number":13},{"id":"6f46ff8ac7fcbc03d865f5423eff0946bdf9351e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-31 04:50:23.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/452/.","accounts_in_message":[],"_revision_number":13},{"id":"da536faf172f5351acb1c5774088dec32678c064","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-31 05:13:16.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":13},{"id":"475d61928ccc147b51a55e1283b3629f09f57e55","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-31 05:21:07.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ee4b8a23c0fe470f8f7bc2a834f11c0a : SUCCESS in 1h 07m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6d1f60e2721a4d2fa07cea92c18bd954 : SUCCESS in 1h 19m 41s","accounts_in_message":[],"_revision_number":13},{"id":"a3c0f5bec2c001a6108ac9ed08041a470990e49d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-31 05:40:49.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5540b2ab6562407695c9d849109ec24a\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03ab0f69c278448c867dd2fa0f1b20cf : SUCCESS in 45m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dbd2b73cfe0f4c7e91513d5d482996e8 : SUCCESS in 37m 34s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b1553b59bff042f2877341b3bec5ca28 : FAILURE in 1h 41m 13s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e251d8dac4f145c2bc7d3ad650aa9d6d : SUCCESS in 34m 45s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"232e4b50834dbc87045f15d675a413251357b254","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-31 05:52:24.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1040/.","accounts_in_message":[],"_revision_number":13},{"id":"1977243ce79461b6a739cc07169b596ee7f0eaf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-31 05:53:16.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e56c896c9494c898dc8c4e9488f5d95\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57d580e9acc54c8d907090f5d1c3e0ec : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d03805a0527044c1a04b7dec14dd2cda : SUCCESS in 11m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff9a78a950f54ffea55ca52200a06eb0 : SUCCESS in 8m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed735c987fab4bc5a7be18e357d709ec : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d3765f2d2ee46408b43c4a5c12274a2 : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/daaebb7da8d24eeabaacbc71dee04f9e : SUCCESS in 1h 02m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e680f9cf300b448aae419417862c8c6d : SUCCESS in 1h 03m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dff49f018ad941c58a216ce069b53588 : SUCCESS in 44m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/140200f616284c16a7a68797bce3cbd0 : SUCCESS in 18m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/08a2c87e734f45619da336a37d1afcc1 : SUCCESS in 13m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/61a79fe6945e4a76b7959de495b87816 : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04ec137be772443aa43cc444c73d3339 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f19c04f68c3147ba8b3b1c41b0feee13 : SUCCESS in 6m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e7594d2b418346b5a3fa0747a0f4358c : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/239ba4052e514a8dbeccaa85856eabb9 : FAILURE in 23m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8240e0e7d5f14aea98712a6d8d8f6781 : SUCCESS in 4m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6ae766bd8957435e98bd8469890dec37 : SUCCESS in 1h 48m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5dec08fc1daf471f9e9ba184b92d08cc : FAILURE in 1h 50m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/da7d199d349e42ecb889459782e35e8b : SUCCESS in 1h 48m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/74c3c7e8f8f84500b60f670dea9c90f2 : FAILURE in 29m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a8a86ec785ce4c0eb24e6795220670cd : SUCCESS in 33m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91413a6313e244c290a9945d28abcdca : SUCCESS in 1h 07m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac3209bfcb5a4535886c57bcd590e462 : SUCCESS in 1h 36m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c92680ac5d014902acbe372bb3cf1568 : SUCCESS in 49m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/161f4ace99e4481ab3a0a0c0d1d20c00 : SUCCESS in 1h 15m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/12e9d5d5574d4af88bb6d092902b7274 : FAILURE in 27m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c68285cb9324d36836b7763281818ae : SUCCESS in 1h 37m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43677e0025f24229ab319f2cb7b2f4ec : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":13},{"id":"d87b4f16aacb3dc98cedfadd0ad9fe1fcac648d3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-31 06:42:29.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/13/upstream-check/cinder-tempest-plugin-ontap-iscsi/3a335d1/ : SUCCESS in 47m 48s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/13/upstream-check/cinder-tempest-plugin-ontap-nfs/3dd0f5b/ : SUCCESS in 47m 19s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"c551fd57b4ecc90c3e4197452cb323b03273debc","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-31 07:04:16.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058 : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":13},{"id":"01cb2badcfee432288d4edb7e6bda3aa61d93eb7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-31 07:48:10.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/13/check/DellEMC_PowerStore_iSCSI/d86bb24/DellEMC_PowerStore_iSCSI/1690 : SUCCESS in 1h 16m 00s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/13/check/DellEMC_PowerStore_FC/708fec0/DellEMC_PowerStore_FC/1888 : SUCCESS in 1h 05m 43s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/13/check/DellEMC_PowerStore_NFS/626d15d/DellEMC_PowerStore_NFS/1383 : SUCCESS in 1h 20m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/13/check/DellEMC_PowerStore_NVMeTCP/66b1ffb/DellEMC_PowerStore_NVMeTCP/1243 : SUCCESS in 1h 54m 32s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/58/912058/13/check/DellEMC_PowerStore_NVMeFC/3fa2ec6/DellEMC_PowerStore_NVMeFC/207 : SUCCESS in 1h 09m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"f2ed8fe3aef2bc17610df45062dc10a0654aaa49","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-31 09:32:20.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1e65cc26ca59bac14999b4e8f193a569ba58ac51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-31 09:49:41.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"5a82e86fcbbac65f9c966affd38037044cefd3cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-31 09:49:52.000000000","message":"Patch Set 14:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/448/","accounts_in_message":[],"_revision_number":14},{"id":"c1556ac68e20a8b637a0caff0f33505db47fbd1e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-31 09:54:41.000000000","message":"Patch Set 14:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/14/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/14/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":14},{"id":"90bf4864051b128975452e6ca6b9e052c07887b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-31 10:15:21.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f46f9255263743b397753a5c74fb6247\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f856f81e78c8490dabec81c7034c8112 : FAILURE in 23m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b0959b3d77654db2b4e830cd17f746af : FAILURE in 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4626e75ac9216f5dce0da02fdd9b2ff414c9858b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-31 10:30:00.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : FAILURE in 38m 38s","accounts_in_message":[],"_revision_number":14},{"id":"f3ea6b488bdafcc51106ff2e0061991b6425d113","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-31 10:41:45.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/453/.","accounts_in_message":[],"_revision_number":14},{"id":"034f21342bd2a1ea8ab5039ced3ecf73271d4ff0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-31 10:49:42.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":14},{"id":"fd96837929e070b2ea458a66ef7b6720d7325bf2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-31 11:14:56.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1041/.","accounts_in_message":[],"_revision_number":14},{"id":"311950b21cdf7eb31169360a3f82ad6051c16513","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-31 11:16:00.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cb7c3580e4a84b0e97f50364c306f3e0 : SUCCESS in 1h 09m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ec24c68b532543b7acaaf46156ec70c7 : SUCCESS in 1h 24m 22s","accounts_in_message":[],"_revision_number":14},{"id":"58f2a2288138c2c16e6cc9c16b41647b644d629f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-31 11:41:37.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8ec631e9f8964dc5999449769b62a892\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/263d8e2d57db4f03987e28fdca0b8927 : SUCCESS in 41m 50s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9426080ac64c4d7fb870ba8955471fda : SUCCESS in 47m 12s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c35b760d5db24d329f27ff37be4c3de6 : FAILURE in 1h 51m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3227f9dc5c0e4e52b54b44d1a221ffbd : SUCCESS in 42m 20s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"b0ebd57f22936ab11d2ac3f9778a69167e69467b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-31 11:45:30.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ae652a6baa4f7a9236c41fb3a3ee56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1a4f52176049509a8ffb6e33132158 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18e431e38a2e4f36851b6bb8018ec376 : SUCCESS in 11m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0962b2e5aad348329900228041c2d4bc : SUCCESS in 12m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e3085d4032f4477a2f38a2406967b69 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ada5726391f4440b3e58e2bf2ced2bb : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/deb7bd16d3634be190ac030a16b6b828 : SUCCESS in 1h 02m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0032d3386582438fbe175f72a86c0ffb : SUCCESS in 1h 27m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/caded31d4d1e4de1b06d7c8af5d7b66e : SUCCESS in 58m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a60e0309a9784a5f9ac6d4d43f44349c : SUCCESS in 17m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea4fbb8b21fc4130a77b6b32b88b8480 : SUCCESS in 16m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/295bd10ab89d40d58f2f276f5d46d42b : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7aef608bd9b4422c978e9a118d4bd006 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c84f399f03344a35adbe03e9a56fdc02 : SUCCESS in 5m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/719b10bec40744f29fcf7af2ea27a79a : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/904eecd3ea154cbb93ed62e63e5d69c2 : FAILURE in 25m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38e832395cf54a078dceafd742558583 : SUCCESS in 5m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/36a957dc6e3a4cbaa2c1bd2e1202ff17 : SUCCESS in 1h 36m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1cf7776ea71c4698b1d7eb4cd53c0e30 : FAILURE in 1h 54m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/56482d4ec9134e1992598e724fc4901e : SUCCESS in 1h 17m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/754eaa510f254b03b0216d844a0c0fbf : FAILURE in 13m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/646ef20618ae428ea2e31a1f780e949e : SUCCESS in 36m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c65be4617a594591a3de86315de2e28e : SUCCESS in 47m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c36d88013f164e188197c71a997e7815 : SUCCESS in 1h 41m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/996ecc4e430f43cbbf53ac49d2d48daa : SUCCESS in 44m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e06202c43681404e800ace0ea3c843af : SUCCESS in 1h 05m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/aee543a560d04cab82e572d924ecd6fd : FAILURE in 29m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/42b209dfe30247e6920456d92b7e03b8 : SUCCESS in 1h 42m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb999a8b5bc34fcc8d3ebdd0b0626c4e : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":14},{"id":"10a65b28e732308f73ba589b537eb0cc22265a50","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-31 11:58:14.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_3.6/bab1cbb/DellEMC_PowerFlex_3.6/2017 : SUCCESS in 2h 08m 12s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_4.x/af96d93/DellEMC_PowerFlex_4.x/899 : SUCCESS in 1h 44m 32s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_4.x_NVMeTCP/ee27625/DellEMC_PowerFlex_4.x_NVMeTCP/864 : ABORTED in 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"b35ca54ef54dcd23d00c5e9054ef916e5c193f4f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-31 12:00:08.000000000","message":"Patch Set 14:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/14/fc : SUCCESS in 0h 42m 12s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/14/iscsi : SUCCESS in 0h 56m 02s","accounts_in_message":[],"_revision_number":14},{"id":"d9c4ae4bf1db4ee876fe5bb869fe04c8f450d56c","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-31 12:15:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"56ca725ae2b01aaf56f92b01d2915c78f92c9a94","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-31 12:35:44.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/912058/14/upstream-check/cinder-tempest-plugin-ontap-iscsi/5c4b482/ : SUCCESS in 49m 17s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"9e92eb5cb1fcdd6130cd40b8abda216eedf734bb","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-31 12:51:42.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":14},{"id":"da81fe1cdbc4ddc187ed2aadc5db40f23db3f59a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-31 13:33:34.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerStore_iSCSI/a9866fa/DellEMC_PowerStore_iSCSI/1691 : SUCCESS in 1h 18m 52s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerStore_FC/4c2bcb0/DellEMC_PowerStore_FC/1889 : SUCCESS in 1h 24m 15s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerStore_NFS/8d34fd0/DellEMC_PowerStore_NFS/1384 : SUCCESS in 1h 14m 29s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerStore_NVMeTCP/fa6cbc8/DellEMC_PowerStore_NVMeTCP/1244 : SUCCESS in 1h 47m 40s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerStore_NVMeFC/7920b53/DellEMC_PowerStore_NVMeFC/208 : SUCCESS in 1h 09m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":14},{"id":"12d586004c685bbdfe2caa9dc8ac4ecd0b036bd1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-31 14:52:16.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_3.6/13d45d2/DellEMC_PowerFlex_3.6/2018 : SUCCESS in 2h 05m 19s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_4.x/d0c009e/DellEMC_PowerFlex_4.x/900 : SUCCESS in 2h 05m 05s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_4.x_NVMeTCP/1640865/DellEMC_PowerFlex_4.x_NVMeTCP/866 : FAILURE in 2h 36m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"86e7933bf48bd156f9317c8cbe34123e033f17de","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-31 15:33:27.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"214a6428150910eb36073cce67ba93e47da31554","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-31 18:09:07.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_3.6/b583a4e/DellEMC_PowerFlex_3.6/2019 : SUCCESS in 2h 33m 42s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_4.x/c0f6652/DellEMC_PowerFlex_4.x/901 : SUCCESS in 2h 08m 21s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_4.x_NVMeTCP/094c252/DellEMC_PowerFlex_4.x_NVMeTCP/867 : SUCCESS in 2h 35m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"e6c615b1220caf4b603fbabf3f3fb268b01c73e0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-01 23:04:32.000000000","message":"Patch Set 14:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/14/20240901231412/NEC-V-ISCSI : SUCCESS in 4h 14m 28s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/14/20240901231412/NEC-V-FC : FAILURE in 8h 50m 15s","accounts_in_message":[],"_revision_number":14},{"id":"d6f9d25966c7e0dc2cced27d3c0ca8d9fc5b86a2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-02 08:55:03.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/14/20240902092350/NEC-FC : SUCCESS in 5h 45m 52s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/14/20240902092350/NEC-ISCSI : SUCCESS in 8h 31m 10s","accounts_in_message":[],"_revision_number":14},{"id":"268630976a1d4059135a8c72a832216a679e2c77","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-09-04 10:41:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1261a46de6c1b5fccddd17bfc3ce5cd23ca1aab7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-08 14:13:07.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/912058-13/HPMSA-iSCSI/ : SUCCESS in 3:18:26\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/912058-13/HPMSA-FC/ : FAILURE in 42:52.49","accounts_in_message":[],"_revision_number":13},{"id":"e5605b034f26e8df4c495d47cd79a57476568089","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-08 23:18:00.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/912058-14/HPMSA-iSCSI/ : SUCCESS in 3:20:09\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/912058-14/HPMSA-FC/ : FAILURE in 43:31.81","accounts_in_message":[],"_revision_number":14},{"id":"c31380110eaa9cf9790c5320a6851ff571e3e1f1","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-09-09 01:02:14.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/912058-13/Lenovo-iSCSI/ : SUCCESS in 2:25:32\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/912058-13/Lenovo-FC/ : SUCCESS in 2:16:29","accounts_in_message":[],"_revision_number":13},{"id":"518ccdb205f828d833cf031ae9cd9ebc784db6dc","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-09-09 11:48:11.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/912058-14/Lenovo-iSCSI/ : SUCCESS in 2:25:06\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/912058-14/Lenovo-FC/ : SUCCESS in 2:17:59","accounts_in_message":[],"_revision_number":14},{"id":"ba8e2d7aedad4167099fab136ca756f7cc012662","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-09-09 17:42:34.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/912058-13/Seagate-iSCSI/ : SUCCESS in 2:23:07\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/912058-13/Seagate-FC/ : FAILURE in 2:25:14","accounts_in_message":[],"_revision_number":13},{"id":"da8084eef78fe991c0998784fd99623f3d4f43ab","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-09-10 01:49:39.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/912058-14/Seagate-iSCSI/ : SUCCESS in 2:24:35\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/912058-14/Seagate-FC/ : SUCCESS in 2:17:21","accounts_in_message":[],"_revision_number":14},{"id":"d18c286c1ce126dbed92f3f0bd427d82aba16f2f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-12-06 09:45:40.000000000","message":"Patch Set 14: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d835557c12b41239667c6a8fea0e32e05be303ce","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-12-16 13:17:37.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0a3e74fa20f0490e5f2bb5ae1650368a3c82b1ce","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-17 19:23:30.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_3.6/04bbda6/DellEMC_PowerFlex_3.6/2589 : FAILURE in 2h 29m 39s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_4.x/ce99446/DellEMC_PowerFlex_4.x/1428 : FAILURE in 1h 56m 35s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/58/912058/14/check/DellEMC_PowerFlex_4.x_NVMeTCP/101e830/DellEMC_PowerFlex_4.x_NVMeTCP/1304 : FAILURE in 3h 10m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"b44c8c7f65fcd07d1689f0d671452be6b4b07c3b","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-12-30 10:17:27.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"cda7aa0d555080b20006b4c8755bfe46f9b1d3b5","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-12-31 09:39:19.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"41eb1dd2db8071cbc4a91281f0474517a48122ff","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-12-31 09:45:16.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a0f3ca83792b6c2262a0930226ebb7bcd34d2894","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-01-02 13:07:29.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a90d5ed1947e209fab3c2ac5c764a04a9d64ab6b","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-01-02 13:44:18.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c887b2f4b12d4da2633afa6bbdd1b1986514ebea","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-01-02 13:46:52.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d2c64499d3f2fd32d27f49771abf0107fa718861","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-01-02 13:51:38.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"41c2013d167b2c422d91bd0668c01b6a76798d99","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-01-07 09:12:06.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"4a229a94c10e0055debeb415810f548bf98c1fac","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2025-01-07 13:48:26.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d52fe5cd6b247f799b085637cf5309dece9a43b6","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-01-08 20:24:52.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"86c7080658f27b849f7951cbca2d908f5dfbf228","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-01-10 13:55:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"cac57c2b3d277f5c5f9b5766cc86723ec611c674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-01-26 04:49:28.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"a9389b0b909930d2a5f20e54eaa824d6dc59c108","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 05:13:24.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e63a7b4446c2421dba7ff77db355430b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fd64e59526042abac8d95c7344ea16a : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2f85679919b439eba8ab2550fa0d8bc : SUCCESS in 22m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"552df7de6e9cebb22c8e23d5c49c9250dbbd0630","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-26 05:34:06.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":15},{"id":"b7c9a1ce832363a53b0e4ba3b5967fb6d044ab45","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-26 05:57:44.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 06m 51s","accounts_in_message":[],"_revision_number":15},{"id":"e737f4c3a7374b5cd6d4a057f4670846f029bbbd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-26 06:01:20.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a5ecacf7835b47b5a94297038d87a820 : SUCCESS in 1h 09m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9361c00d8fa541809f22a7731a467412 : RETRY_LIMIT in 8m 09s","accounts_in_message":[],"_revision_number":15},{"id":"4759cd3e3367f1015c11e748cf77fbf697626969","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-26 06:32:44.000000000","message":"Patch Set 15:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/51e2eed8ca0347a781f434c7ec71e86e\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d086413fdd4445b79a564f25a30455f9 : SUCCESS in 40m 22s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c507fc184ebf4526b6b71fdbe36f7a3a : SUCCESS in 43m 46s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d3d17386e064d9fa97c09495225046a : FAILURE in 1h 42m 44s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/85a52bae14304d909cc5c34e1664200b : SUCCESS in 37m 45s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5323bc71c0624dfaa1b7d46e0290bb20 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c7b4568d48140b7841f2620c3064eaf : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/30b423c77331417e891a290a1f9c905c : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/41ff93fc83bb457ebd2aa71895dbf6cb : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":15},{"id":"b67751c134ba4c14786bac0721ffa457aa366e59","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-26 06:53:54.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/912058/15/ : FAILURE in 1h 56m 04s","accounts_in_message":[],"_revision_number":15},{"id":"0b879ef22a374d626e99a2098fbd6113e75f1cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 07:43:18.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5eee091259fa4c16a6c0b297dd0ae570\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/079026d40b2f457aa62a51d8293d7082 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b88ade7555d34333a342798248f935ce : SUCCESS in 9m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdd41526c8314b3fa60319b4f319517a : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a2972492be4fd3a44be04a4686483f : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/2bcc4bca6cd94942aa7ac3662b906048 : SUCCESS in 41m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f34d15a882ad49bd915dc2bf98f47814 : SUCCESS in 36m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acc1d09098284aee9c6f818dd8fdbcac : SUCCESS in 1h 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53bf5f02a30a4058b91ea967eb5dce96 : SUCCESS in 39m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88e93f92689d4ab4a56dad619e522391 : SUCCESS in 12m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90b0407a0e6a47ca884f9cb782f0e501 : SUCCESS in 19m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fb4257869cf248c8bd287cc9431d313b : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aa30ca66e7154d568c712d719f33cbd1 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42c2cff340464fb29b4671897c68d7b4 : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9cd8c34a88ea426b98ba8dea89dc5af6 : SUCCESS in 5m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6dbc2662da3e41f280c8b2f241ab7dce : FAILURE in 18m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5fb65b37985d4e4fb183c884b8e0390d : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/80ac65aa7f0b481b9d6245d8ce4df2df : FAILURE in 1h 27m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/abd8c31f6ee94659a3339646be1bc514 : SUCCESS in 2h 46m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/98212fb633744dd283870cf140ad0253 : SUCCESS in 1h 22m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1773dcbdd35a492a85a9fdc8e6dad1cf : SUCCESS in 1h 30m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dfd5defe31d54b1e9f37819a2ee40f6c : SUCCESS in 22m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a6a089e89aa540cf88cc0e211b4fcfbe : SUCCESS in 50m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b28d7e231f5441da850e70df1c978e10 : SUCCESS in 1h 14m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/05cdb9552739406ba79e930b38a2bba7 : SUCCESS in 30m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c8128331120b4c8dbc406311d0a2da82 : POST_FAILURE in 2h 28m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/09eadbc520c746ac84285c3a909b213f : SUCCESS in 1h 10m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a359bf1b5cc541bd8fd7c1aa38e1a6ed : SUCCESS in 1h 12m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c9f54f604834b28a2c6a332f1cfafa1 : SUCCESS in 1h 09m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cab7fc8351e4183a98b2401dfab356c : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":15},{"id":"365625e5cd59ae8a1a05a121523c901f04717a31","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-01-26 08:22:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5bb64926a191e9cbd45897c88a73dfd127bb66b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 08:42:40.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c51deef32a974fdfa5db5c0ed254ffa0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75796064280d42a0b4ffc47184b353f3 : SUCCESS in 17m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07959c7d1a964e84908c009c00681c70 : SUCCESS in 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"bedb6ec2096627aeab2e2a72ded1183f6769a24e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-26 09:14:51.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/15/20250126134932/NEC-V-ISCSI : SUCCESS in 4h 02m 15s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/15/20250126134932/NEC-V-FC : SUCCESS in 4h 25m 17s","accounts_in_message":[],"_revision_number":15},{"id":"ce05ca030df78fd4f687095b953fc114a62903cb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-26 09:27:41.000000000","message":"Patch Set 15:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-15/FC/ : FAILURE in 4h 37m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-15/iSCSI/ : UNSTABLE in 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":15},{"id":"516c481785ab5e1d443caf51c61ffb94662e4efa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-26 09:41:06.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c206441839514fd09164d6ee45d02cb9 : SUCCESS in 1h 16m 59s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0bbd21f1d3fe49f5a9076d9d47a64e31 : RETRY_LIMIT in 7m 31s","accounts_in_message":[],"_revision_number":15},{"id":"ba6f8b230492d3617194cc940a2f1207a7b1c4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 10:29:20.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41132ad9789c43b78cbd001c60a6d83f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7b57f5a113d4ea6a08838f0e9de66b3 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c74ce0480282480bbeaf4d9f41a4c924 : SUCCESS in 10m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/770e7317c8cb4ad4a908ae6e9497ad31 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8eceb8be7b64d69b0c876de384df793 : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/1939b009c98e4e16b7783d5320ebc01a : SUCCESS in 45m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79f11ae936c54dba91d7a4f3b880e91d : SUCCESS in 45m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8c3a19f10414ec7b69a2d4e4fe3c62f : SUCCESS in 1h 15m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90a8d9f060f94a9b9c0ed3e215826f41 : SUCCESS in 38m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f660754558524723bcaa698cc77ab440 : SUCCESS in 14m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb7ecd7f470240c6b1ade9db322072e8 : SUCCESS in 21m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/962b3d5e939243e7af67b2da605ef51a : SUCCESS in 3m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/43074f3c85b443abbfe1185bef0774ad : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f91897e85df24cf98f0261b49597d528 : SUCCESS in 6m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ddbfd4dc432748b5add021e17a568fb6 : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0cd41db1a5a345fca931011ed6e41e88 : FAILURE in 17m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e7dc8c11860436a9b804a77bd50de5f : SUCCESS in 7m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/067c5ebc340949c495dce3f060243fec : FAILURE in 1h 21m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/857126895eeb47b09c3bdd46161edc88 : FAILURE in 1h 57m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/70fbd5a4c811467f82c30d81007f786f : SUCCESS in 1h 30m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/96d9138294d94740a4b2443fab8cd7b9 : SUCCESS in 1h 25m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/337844492af64ad883fbe9dbe3199d19 : SUCCESS in 24m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b5e6990370da48cc8b851b6a4d6e8f35 : SUCCESS in 54m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d09c3cc6467048f9a55fc3b2f0ab8743 : SUCCESS in 1h 24m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d395ef9a14164928891e69f9b930a41c : SUCCESS in 32m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9edd8e7898d040cb9663ce5efa1441f2 : SUCCESS in 57m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a372b21e93454e359e031e7990dffbeb : SUCCESS in 1h 11m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f0394137840841bea0960066c0e2ca40 : SUCCESS in 1h 19m 42s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ad4842d838a43d68a57767d2a0492ef : SUCCESS in 1h 12m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4130074476d84bf2b6f33c2dc6093135 : SUCCESS in 35m 58s","accounts_in_message":[],"_revision_number":15},{"id":"2392b6e982864d4852b621143ebea26f81329887","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-26 15:25:12.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/298/.","accounts_in_message":[],"_revision_number":15},{"id":"8451740d8e9b0cd4df41caf8ed121e763c4236c7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-26 15:48:20.000000000","message":"Patch Set 15:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/280/.","accounts_in_message":[],"_revision_number":15},{"id":"743e9e6664bc16db4a47b5bd91aa3a25a94d8435","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-27 06:32:06.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/15/20250127093605/NEC-ISCSI : SUCCESS in 5h 30m 38s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/912058/15/20250127093605/NEC-FC : SUCCESS in 5h 55m 59s","accounts_in_message":[],"_revision_number":15},{"id":"0e869cdc691b71b2ab546c08ad11a0ef051d83ae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-27 14:07:15.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c994bdd42348472aaa8c07d0dd86d4ff : SUCCESS in 1h 08m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/474589e626e842ee86a26da30a60816e : RETRY_LIMIT in 6m 54s","accounts_in_message":[],"_revision_number":15},{"id":"615df59f887378579f6683d598b58ec2c3a351fe","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-01-27 15:07:09.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2d45561265c835b80641e031c38a7f400693624f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 15:31:12.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad9854f1a721488eacd37c803287e5fd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17f91969fbb94eb2a1c269aa2711f808 : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a2b981ac5874c97b388885725778353 : SUCCESS in 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"28687466ec86de5eadd1f57c3d5126c259404636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 18:22:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca6afc65f3534c3c889483d64bd546d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbbc3009d4dc42a185912c623c5c210a : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a906123a214467b96636b31b55399f1 : SUCCESS in 11m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/892d30df6e9f4c56803c1301c5680620 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a81e49326314b04a92b25058a791486 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/07dfb2ba84e5474e9c535d12b7d1b0ca : SUCCESS in 36m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdf77a34aa0248a6864b0a88433bf5e9 : SUCCESS in 36m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/437b00cdfe8140cea893c2ba1d586a17 : SUCCESS in 1h 31m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/019203ec841b44178248378f215c48bc : SUCCESS in 1h 04m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b24306e727b94a198afe63d70928bdbd : SUCCESS in 19m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/77b5cc0b85024b2882daf15e1552b5ba : TIMED_OUT in 41m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da68b2064705425c8d8c88008bf9ffff : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d4c784e015a470380a59d44cd33fc35 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea1fc467994b48a0a40e5ef4b8959b51 : SUCCESS in 8m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f768dacc7e7e43e2bbd15ddd802f6c67 : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b25b2a794f5f4afe9803d32314a0d72a : FAILURE in 37m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/30382b6fb6554b7ca566a24e5bc083be : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2687d576c3824a7eaabe00cd3a59f5a4 : SUCCESS in 2h 03m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1e7832518efc4a7cbe97dd95b7abfb69 : SUCCESS in 2h 41m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fb361c19e40a4975bac96185f2ac196a : SUCCESS in 1h 50m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/76c4619ada7443acb8098d64f802c1a4 : TIMED_OUT in 3h 05m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c5e488d0484843a5b3b427895e76e65a : SUCCESS in 40m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8e834091e44d4bc9b6e0630a97f3d9d9 : SUCCESS in 1h 15m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e6e1d80538c4cda810a76b619cf25b8 : SUCCESS in 1h 35m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8f55803cbfa94efe8d86821b3a1a3abb : SUCCESS in 42m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b2231662161453391466381169d427d : SUCCESS in 1h 00m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7dd182d6e98b42c2b24e0b7be7571e40 : SUCCESS in 1h 52m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f78225682f7747a3992890f16c524864 : SUCCESS in 1h 43m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5df7a29562714872a472126aa0f31220 : SUCCESS in 1h 27m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d768ac53c7894567b1d581599adce9a1 : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":15},{"id":"cd353ba51ce106f438d9c33ab0877345faa38fa9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-01-27 18:23:33.000000000","message":"Patch Set 15:\n\nBuild failed.","accounts_in_message":[],"_revision_number":15},{"id":"5bfe7c70d9281b38c85004bd58bf442c6d0b8234","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-01-27 18:23:42.000000000","message":"Patch Set 15:\n\nBuild failed.","accounts_in_message":[],"_revision_number":15},{"id":"6ec2cd74064313ecc3a8af2780dd3dded5160205","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-01-27 18:24:23.000000000","message":"Patch Set 15:\n\nBuild failed.","accounts_in_message":[],"_revision_number":15},{"id":"cecafb073ed7b4223710158370a75402e2bc177f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-01-27 18:24:32.000000000","message":"Patch Set 15:\n\nBuild failed.","accounts_in_message":[],"_revision_number":15},{"id":"35c0c1eccccf9db504aa10b4c597f62064d610bb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-28 01:02:51.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/56eb2cbed5884131808e116913bf7951 : SUCCESS in 1h 16m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7e5b777898034bbc99ac147f982f3e5b : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":15},{"id":"34df4332c1e14ee4dcd2fa76119e21ea8fe98d3f","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-01-28 04:13:58.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c72064f98e2bba75c2f60ef3b9ec6217bbf79028","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-28 12:03:10.000000000","message":"Patch Set 15:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/28_01_2025_11_57_02\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":15},{"id":"49fc6793d7ee81f8f060e481d972c5c6d447aee8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:52:19.000000000","message":"Patch Set 15:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1270/","accounts_in_message":[],"_revision_number":15},{"id":"d650bca2cabc08d1436e01a0289095ee2a80b92f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:52:21.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/15/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912058/15/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":15},{"id":"429e3cf1c5f36939a8bbb926af8aeefcf9726428","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-03 17:06:19.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058 : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":15},{"id":"6404856448c0a7f001d602d293595d0fca4e7ba5","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 08:05:52.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f26fde65711abb7dd327a168ca87fbb2d192aa8a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 08:33:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e4f93074c183ca130de2633d5f1594ab28770678","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 10:54:22.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/55358d3172ce42ed89d257213de19f82\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/dd39b054406f4246a488c086c7bdef17 : FAILURE in 2h 20m 11s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/7b30ef1e4e334e25802e43cdb464c3e6 : FAILURE in 2h 17m 59s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"13c55187663f77df2589a9201043a031918632ef","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 11:07:43.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6ad2cc2a9a0bce6e83923c752f68140407d476a0","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 14:10:49.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/899c6233b6094b05b8fc27d2f96988f8\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/ebacad53d8854bcd90d24402bef48fac : TIMED_OUT in 3h 02m 38s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/878cce6d82284098891544ca679e9cf6 : TIMED_OUT in 3h 02m 43s\n- powerflex-nvme-cinder-tempest http://localhost:9000/t/openstack/build/4deea8c5ce1045ffbb61625e726a5c06 : FAILURE in 59m 08s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"6ec2ec6d028d57fffc8f09fe1aa346c58d5a4194","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 14:15:09.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ab7a5ef2766d77919d438cef469b9693c02967e4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 16:53:11.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/7e32b08dde464df4b8b4648f9c1b5fb1\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/d853bad31af1488dbad864cb418dac90 : FAILURE in 2h 13m 13s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/5f43d9e948714313a4e23099cbdb818b : SUCCESS in 2h 37m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"1f630f234d3f6bd16d8ebc888604eaf23b3f42ac","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 17:41:05.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"8eaa7e9fe6cc94b21c84d80dc3f5fbc19ae0aed1","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 20:24:17.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/1591bcbdaf7f4ee09fc4607943a10eda\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/3c0668778e214640be94912e6d3a4ce8 : FAILURE in 13m 44s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/6ea6736929494154a22679f463c42736 : FAILURE in 2h 42m 49s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/2a0a3b859b7c4ca291994ee450251756 : FAILURE in 1h 03m 50s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"b3992df52e980516505562df1a09c99fbcf13caa","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-10 16:40:23.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3269a63b24dddb2e7e565029cd46e926232dab92","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-10 16:40:58.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/91c6978f807245dfa6c9ed7290522d9c\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/d7b304bc81ad44b58a0ace22f3d3a91d : NODE_FAILURE Node(set) request 200-0000288644 failed in 0s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/2abfbb63ee2a48e485cc8399f2affc2d : NODE_FAILURE Node(set) request 200-0000288645 failed in 0s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/5a4893bdc40843b98a43db4fd575de08 : NODE_FAILURE Node(set) request 200-0000288646 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"c54c860656914bb611833c426dd98ddf3b453e2e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-10 19:25:33.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/b36e67ccda6549a2ba400c1fa57371df\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/0861999d47834fb5a293248bfc117ddc : SUCCESS in 2h 44m 35s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/8074135ddf26477b84734e0b61843b0e : FAILURE in 2h 38m 28s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/9dde84e4059b458c9fb4fd65701f0120 : FAILURE in 1h 04m 19s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"dbe1a50897089e3b17795f0aa703e66561cec060","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-11 15:46:53.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2a887c0d34ad7c9e77480028a49f730402db0043","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-11 15:47:23.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/32c98d9c1d23446c919e4c659cb87272\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/25bd9676b0aa4642bcb3fc39bf0d897e : NODE_FAILURE Node(set) request 200-0000288647 failed in 0s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/3020ae4a192c4aa8bac8f5a760ac1d13 : NODE_FAILURE Node(set) request 200-0000288648 failed in 0s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/b6756c92d99241718c31c494eea7981b : NODE_FAILURE Node(set) request 200-0000288649 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"a07fcb09bfefa5f631d705c6067e44b63ad48e83","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-11 18:15:00.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/dac223abf78b4323a538977c5b3ba5c7\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/b57428e731454bd4b8ff43ea51134d35 : FAILURE in 2h 27m 38s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/84c737a95877477298bc00e887b20f33 : SUCCESS in 2h 21m 01s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/a42ca46fbcdd49eea513823b8de91477 : FAILURE in 1h 00m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"38fa24d1f5d32a71c9b95ade46a75da19ab014d4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-12 08:20:42.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"83889930ad0119c8d138403ad0521814a17db2f2","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-12 08:21:08.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/894909a037d04d2aa4f0009f2dcbdd34\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/e4b240b6189f4251a1c564a306fafc3c : NODE_FAILURE Node(set) request 200-0000288650 failed in 0s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/8d405f4c45f149aa9ad038d9f4e9ade0 : NODE_FAILURE Node(set) request 200-0000288651 failed in 0s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/335efb2ca65e4dbc8f4f7d1779e26738 : NODE_FAILURE Node(set) request 200-0000288652 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"fa199ad8fb5e1005d81bd21acd897dc2c2b64ca5","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-12 10:56:31.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/4c199b48eef9403e884ce28b610207a1\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/d40a386d11b241feb047b6356f127c54 : SUCCESS in 2h 33m 31s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/8e7fa8156d7f4b2398febe27dc93daee : SUCCESS in 2h 35m 22s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/f2f8f23e30ec43748246c0676669ec64 : FAILURE in 57m 16s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"3fa3fa48ba3b696f9885efce5f12f7b7b656da19","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-12 14:46:31.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4c45a136ecb785ebca9c98dc906bb3f592398c3d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-12 17:16:53.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/2361c3073bd94ccfa8bf1cc06cffd942\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/133364f7fe1f4f7cbfb7a2109f67c130 : FAILURE in 2h 29m 56s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/0ede64560af14fe8b6e08886aecef337 : SUCCESS in 2h 28m 37s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/1fa5f48a50fb42568861e3c4f16db4a8 : FAILURE in 1h 01m 59s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"5444c522c4dc57fd227ce6d5caed6fcb4078046b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-13 06:46:54.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"399b570e8ce6b14aeb531496a0d66bb8f075fbce","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-13 09:17:42.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/9161f2a0da6e4b058a0db71cd6ca60a1\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/7457f0698813453e9a81d9262f1cab7d : FAILURE in 2h 24m 41s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/ce5a17eb68cc43bf85d579c6ab34e5b3 : FAILURE in 2h 30m 17s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/c13ef6fd256e4811aa58c13af2835d64 : FAILURE in 2h 17m 08s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"ce997fb305ba199f2316572f3e911ec012def9ae","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-13 10:03:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"fef825b216b259523611ea62a96d57637a70695e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-13 13:07:44.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/14e718a1fdcb4896b863d23b52f534af\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/10f79419d3ab488bb95b355bc2798905 : SUCCESS in 2h 38m 56s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/95068cd542da45eb92af591f1d3ad56f : SUCCESS in 2h 37m 38s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/3ca1e0df7f3644699fd2ada9401b833b : TIMED_OUT in 3h 03m 53s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"5cd22003e953d6761b5a9feaac36cdf6e6b6c8e3","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-13 14:01:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bd208459785ef906338a559988065e0066b08587","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-13 17:01:58.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/32731c1781dc4018833433dc81e62f47\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/37ed11c55d2f43a18a338b13706b6a0b : SUCCESS in 2h 48m 52s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/876aaa9b47dd429995ec0056bd5bc29e : SUCCESS in 2h 26m 07s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/5dccb8499aa64b97be5bfae4af97a268 : FAILURE in 3h 00m 05s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"d694d63b1783eda4d03417fea12761b699f1940e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-17 13:36:03.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"81639aeaf6ccd751d54c99fdd1a1332979b41d79","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-17 15:14:43.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/476bed0c4221461680d80421592aefad\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/55fd88682d6f4ca0a796886e8ce55518 : RETRY_LIMIT in 20s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/3da452e2fc174d5aa1eb018ab1c23079 : RETRY_LIMIT in 20s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/929a3c81427c4acc9b265f3cd2b3c4e6 : RETRY_LIMIT in 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"de3536cfee3463e77aabdc5b0f16139439d19849","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-17 15:20:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b6bffdca7772e1c473311b1f5cc357ba99ef5a66","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-17 15:22:35.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/402f7d5b79b349189e121176caa0bb35\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/528b8594fa714f9fb14e8097d614da54 : RETRY_LIMIT in 20s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/61dff9a8ddad4c2d87af32e3886621d4 : RETRY_LIMIT in 19s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/7b06dcdd98fc4445a73ae3779a48a268 : RETRY_LIMIT in 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"28b60a37eb5e139b77cd154f4a6d0aa96d320d44","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-17 15:26:09.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"0b447ef8237e6e1f1b66fe61bf2fb5f45c607e24","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-17 15:28:20.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/58261aceda074ab3a76df4468b1da1e3\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/b27757ff7e754ce390b44c19791d1078 : RETRY_LIMIT in 21s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/0a343b2b7c9a4e48a352919ddeb37fab : RETRY_LIMIT in 19s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/c2df20a68e1f4fa78c8991ca2e036493 : RETRY_LIMIT in 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"e049ea233219ecb0d90dd4b00eeee6780b6e037d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-18 09:42:35.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e9f6f81b2219369d9931e8c81443cfa8927df2b7","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-18 12:53:39.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/0d4c559d5f1a4257bfe74faaa1a0ab9f\n\n- powerflex-v4-cinder-tempest http://localhost:9000/t/openstack/build/f72a228b74b949f49f937a0091f4f385 : TIMED_OUT in 3h 05m 05s\n- powerflex-v3-cinder-tempest http://localhost:9000/t/openstack/build/31f3c8f3cce34babaa150669d8dfd795 : FAILURE in 15m 18s\n- powerflex-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/8a0f12fb326444bbb43549bc140b4558 : TIMED_OUT in 3h 05m 06s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"1a7d502659da1575643dbfdb5e4e60d4482101b7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-03-18 13:00:12.000000000","message":"Patch Set 15:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/15/fc : SUCCESS in 0h 49m 33s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/912058/15/iscsi : SUCCESS in 1h 13m 53s","accounts_in_message":[],"_revision_number":15},{"id":"a21f782efc7e8dbd0564077a7f93d0bb706314bd","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-20 11:02:50.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5458ce4d57b63b4b9bdd5fa56efa6707aa922099","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-20 16:25:48.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/333f009945f449d8821f5c82f94e086a\n\n- powerflex-v4-cinder-tempest http://10.228.225.12:9000/t/openstack/build/588850771dba4f11923025e074af088a : SUCCESS in 2h 30m 21s\n- powerflex-v3-cinder-tempest http://10.228.225.12:9000/t/openstack/build/698d4d8057244b9aace45651c0211966 : FAILURE in 3h 00m 27s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/9b854925fa4c47fdabfdcbe82d90c11f : FAILURE in 2h 44m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"0a7b49afdf5aed3429202a88f921fce7cac85d6f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-20 19:43:16.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1479d6ec40703bd6b2e3b5fb095e5f48e5bf3ea9","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-20 23:26:36.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/3a44e4889d114f1984c394ffbfa1f921\n\n- powerflex-v4-cinder-tempest http://10.228.225.12:9000/t/openstack/build/f833c7f724ee4e6ca2c0e56f8836bb2a : SUCCESS in 2h 49m 19s\n- powerflex-v3-cinder-tempest http://10.228.225.12:9000/t/openstack/build/1edf01e032024d6cafe2cae8f9104591 : SUCCESS in 2h 38m 43s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/d382282bf7be4a81b02c1c0fbc084a63 : FAILURE in 3h 42m 53s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"22a374c7b16ca088483fd728538ff5a9a5f8ee57","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-21 09:08:44.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2ca4f739c621a46f9ee4543ff31c0054ddc6522a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-21 11:17:32.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/92fd60a0e1d1471b84262c0b3cd30ce4\n\n- powerflex-v4-cinder-tempest http://10.228.225.12:9000/t/openstack/build/46b5cdc3ccc44b9d86f278461ed6a611 : RETRY_LIMIT in 22s\n- powerflex-v3-cinder-tempest http://10.228.225.12:9000/t/openstack/build/83d4ba3c40434c7cb6bafff86127de6c : RETRY_LIMIT in 20s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/3b7a003e5f0e4f95b160889eebbebf6a : FAILURE in 15m 59s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"4419404752292bc066dee6d96a3db9df7fa7e8af","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-21 13:22:01.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b87d66bec1265ef33510fab36086c46fee704bc8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-21 16:20:51.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/a39467f537c94ae39064f5855632744a\n\n- powerflex-v4-cinder-tempest http://10.228.225.12:9000/t/openstack/build/71ec3b7209f34a538c3f5a07a6956118 : FAILURE in 2h 54m 16s\n- powerflex-v3-cinder-tempest http://10.228.225.12:9000/t/openstack/build/4cdabd14f2d94734ba929ae7da93c44b : FAILURE in 2h 23m 43s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/215c9356558344259d8965c6a32b7c37 : FAILURE in 2h 35m 10s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"6345ea143265d6b6f48c9366430a00c6b64c175e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-21 19:13:04.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/6d67a4c700044180821dd4570dad12f8\n\n- powerflex-v4-cinder-tempest http://10.228.225.12:9000/t/openstack/build/cf44a37f2b9342fd9c0b28163aeb839f : FAILURE in 2h 47m 35s\n- powerflex-v3-cinder-tempest http://10.228.225.12:9000/t/openstack/build/e8c86fca2235452692ef72ff474ae4ed : SUCCESS in 2h 35m 58s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/797a9e026ffe437aa38e4919e8e1c72f : FAILURE in 2h 51m 48s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"e955db49ae00e68dcd439de03a83e9a76025214d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 10:04:55.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f8fa151ae67a8cbdb4324515cb4367ad00b5f6b2","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 13:26:48.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/b6870277bc57413997b4039a604a73de\n\n- powerflex-v4-cinder-tempest http://10.228.225.12:9000/t/openstack/build/013ae56aeaf34b39a2c9fec89de2a4fd : SUCCESS in 2h 58m 20s\n- powerflex-v3-cinder-tempest http://10.228.225.12:9000/t/openstack/build/7f7a9077830a43a7b5e1eb386f015831 : FAILURE in 21m 53s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/2ea2b7eeade14845930edd2922a50133 : FAILURE in 3h 21m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"67b96f3dd148bd84042119d3deab6a6af14ba5a3","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 14:08:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"692ff8fa9c53b7e0ab3c5282340204c8b8184584","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 17:06:13.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/8fde1a1b5b344e51bda6daa9f3158fc0\n\n- powerflex-v4-cinder-tempest http://10.228.225.12:9000/t/openstack/build/6dcbdb71f4604a2894ba78702ce13e7e : SUCCESS in 2h 30m 02s\n- powerflex-v3-cinder-tempest http://10.228.225.12:9000/t/openstack/build/7b72ce6168ad4d2696a5e2463e45068b : FAILURE in 2h 57m 18s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/09b30f8356cb4377aa91333c442b0e36 : SUCCESS in 2h 32m 54s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"677a0ed3b541e1ae149116fa292041ec9444dea8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 20:00:34.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"277e71a5f15f007725e8fc1e93b5f36654dff107","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 23:23:32.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/9654012a65ed4677bdb2ecdcd1c78d1b\n\n- powerflex-v4-cinder-tempest http://10.228.225.12:9000/t/openstack/build/d3ee0fc3119043e4a1a9c30343cbc596 : SUCCESS in 2h 30m 31s\n- powerflex-v3-cinder-tempest http://10.228.225.12:9000/t/openstack/build/29f5d11125544a44a24635c731327f2b : SUCCESS in 2h 47m 47s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/510ff5b4009d447894dd92136cc5b72f : FAILURE in 3h 11m 28s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"ba37329c12ba692c0c50bd386ab9e5bda1423405","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-26 08:44:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"75515f88f855360b987ce82348b4b6dd5caa79db","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-26 10:51:22.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/3a57423b8d5644b1aee6717e1b5b0437\n\n- powerflex-v4-cinder-tempest http://10.228.225.16:9000/t/openstack/build/641549e1eed7438898cbf09b05ef7b32 : FAILURE in 5m 34s\n- powerflex-v3-cinder-tempest http://10.228.225.16:9000/t/openstack/build/8d7006ab965042e0bf180e25c57c7caf : FAILURE in 5m 32s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.16:9000/t/openstack/build/983d46b1394b4617adb52ef878447a73 : FAILURE in 2h 06m 43s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"90f8955cb7dd74fea9bd8bca280d6c6188b99bed","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-26 12:37:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"dd23ac955e64e501f8a7043edb32ac31f45a665f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-26 15:13:39.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/d43b427c00ac4b0fb98f28f476ccf5e0\n\n- powerflex-v4-cinder-tempest http://10.228.225.16:9000/t/openstack/build/38299fbca8b04f27aeb06391df8f96c6 : SUCCESS in 2h 14m 59s\n- powerflex-v3-cinder-tempest http://10.228.225.16:9000/t/openstack/build/10b2ae240a0d4d29963e741f65485c95 : SUCCESS in 2h 05m 36s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.16:9000/t/openstack/build/24b656c5c5f24f0d8a2124bd9707fda7 : FAILURE in 2h 36m 03s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"eee53b126e89a450b0d04d7185ed63779c6bf9dd","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-26 16:44:30.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ea1f63b72ae21c2f4e1a1a5a82ab71cb89a1b9d6","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-26 19:58:05.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/5608320505254a1b83a348b1c4b04972\n\n- powerflex-v4-cinder-tempest http://10.228.225.16:9000/t/openstack/build/536a42fc671649dc9afddf60c7558b68 : SUCCESS in 1h 57m 33s\n- powerflex-v3-cinder-tempest http://10.228.225.16:9000/t/openstack/build/1b273483b8944f30a6cdb1a394ea30d4 : SUCCESS in 2h 03m 41s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.16:9000/t/openstack/build/d1e2bcd5710b49bc8f2bb7c60120a7e5 : SUCCESS in 3h 12m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"305aa9af50d4f422d43d0bb6eb2023a9d56af677","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-10 14:27:39.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"264b006fd37dbee1f5023cc3725be25800a586a8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-10 16:35:03.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/7c77b170d8f74eda9847cf567178d458\n\n- powerflex-v4-cinder-tempest http://10.228.225.16:9000/t/openstack/build/144e4e14c9ae432ab4435fc0cfe22c75 : FAILURE in 1h 39m 13s\n- powerflex-v3-cinder-tempest http://10.228.225.16:9000/t/openstack/build/510ef29dce974bc28cab46df5fc17d9f : FAILURE in 1h 43m 24s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.16:9000/t/openstack/build/90d1e6bd987844ec92b8e6ff49b67f6f : FAILURE in 2h 06m 29s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"ef29961ce7167fcf4cab482ab93abfb03ff3a32e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-10 19:11:43.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3009d3c8eeeccfa09bb2a1efd7d0dd6951debb55","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-10 21:08:34.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/147bc4dc93c74500ac6d781a085f0fcb\n\n- powerflex-v4-cinder-tempest http://10.228.225.16:9000/t/openstack/build/cce19389258848f7ae771bb93780e409 : SUCCESS in 1h 56m 00s\n- powerflex-v3-cinder-tempest http://10.228.225.16:9000/t/openstack/build/bafb71f543404cd49c239788d913f6c2 : SUCCESS in 1h 55m 54s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.16:9000/t/openstack/build/1dcc7aae482841b1a5809a2885a413c6 : SUCCESS in 1h 55m 02s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"58c677a5a0afe72163873dd41586b18f0ef0ca69","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-17 16:18:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b06a498de13725b0629fd5bad78a69c9b3b57928","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-17 18:07:38.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/4daf1ab2d4734ba18d53529bf78ce74b\n\n- powerflex-v4-cinder-tempest http://10.228.225.16:9000/t/openstack/build/818ef79d8c7e4e338aa4d5ce37534d1f : SUCCESS in 1h 48m 44s\n- powerflex-v3-cinder-tempest http://10.228.225.16:9000/t/openstack/build/89a8b3dee11a4b1e973dae3dbc3fd09e : NODE_FAILURE Node(set) request 200-0000113329 failed in 0s\n- powerflex-nvme-tcp-cinder-tempest http://10.228.225.16:9000/t/openstack/build/1665050564dc45aab96f13b20ad42e86 : NODE_FAILURE Node(set) request 200-0000113330 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"6b5cad13f6ae04dcc4d799ee5d6acbe1f31f1d85","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-11 18:19:02.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"7ee4b75b0b007a0f1ed031945b06c2a7ac541860","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-11 20:25:34.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/ca0f990c940a4bcb90115e1a352196eb\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/b3d6718c95ba4f79a3a430cbc684be99 : FAILURE in 1h 12m 36s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/977eadabad8041489cd88479d54b9491 : FAILURE in 6m 20s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0f468acbdd05420cbf578cf112712d75 : SUCCESS in 2h 04m 44s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"ba633fc41ce14a1501d2dffd423b6d6dd1322a29","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-05-13 08:48:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"93a027093b571211585266d285b95016cdaae928","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-05-13 08:53:46.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4d4988a93d3eb724a7b50c9bd26375aa010b508f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-13 09:07:20.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"75c89277f5198ccc7379ca33742ad2953f8962be","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-13 11:36:54.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/6be9af354e694b658561e66d4fdac632\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/43f38bd674564344b26041b2e8ab7174 : FAILURE in 1h 35m 01s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/6cf55894bb85407ba83c17963f5fe923 : SUCCESS in 1h 44m 17s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/fdf1ec918f5549dcab5e02a21569c72c : FAILURE in 7m 12s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"602431573849b3715ce88199cd9cf07567e571a9","author":{"_account_id":37822,"name":"Atanu Sikder","email":"atanu_sikder@dell.com"},"date":"2025-05-14 05:08:12.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c78b8d3b7b8c6e5d951a328ec77efe5646dbe5ab","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-14 07:01:12.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/dccaf32dc70c42acbd7de6c1274b28d7\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/51785552c1c64004bc545ecaf1d61fb2 : FAILURE in 1h 28m 30s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/a6d07e407c13495a882e40faa95554eb : SUCCESS in 1h 34m 48s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/18afa47105444461b550932446a821c1 : SUCCESS in 1h 52m 10s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"a56c7792af67f49fd48fd27cd19c62c184826b83","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-16 04:39:38.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"8ee84c9d1c9bd5d183dd5fc9f9926ea8438f39ff","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-16 06:58:22.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/bcfbdddd33e74b4195174105f92578a6\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/a1a3031004a645f688d5e5ebfeff8e6c : SUCCESS in 1h 48m 13s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/772179e12b24448890dac00976e98aae : SUCCESS in 1h 56m 15s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d6a14771deb7411c9f75927d81f1ea5f : FAILURE in 2h 17m 52s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"51f855a9f79d900ddf377a3f1ec785a01f24faa4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-19 07:22:55.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"7dd47681249326d357043eeb5d30b9e3ae0a3138","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-19 09:06:39.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/161f7c2171204d1b969357cef74113dd\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/3b6f1828c2e74034b70cc8bf0eade9ba : SUCCESS in 1h 43m 19s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0389134ab4e446ca984064d823b3aab2 : SUCCESS in 1h 35m 34s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/fca65801273b41fea7ed524e5f86be81 : NODE_FAILURE Node(set) request 200-0000115835 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"08357f91cdc1251c4e9cf9287cbaad4405801386","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-22 18:56:06.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e9f567a478fa6cd5024ccae81f8d3f1120b5e7e3","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-22 20:34:02.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/bd8d02492c274c28b387bd68beccaf6c\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/9eebc3ac708441979f37e1c29c6dd7c1 : SUCCESS in 1h 37m 33s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/c3b30becfb8d416997fd2da7aa0099bc : SUCCESS in 1h 32m 34s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/8c50df5796804691809cb3f6e427d263 : FAILURE in 1h 16m 52s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"b9439dcd7164e3059d7e99acc2e99e6c7f91c5b4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-29 07:13:02.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"222c284c1fb6a2dcee505283e4c88bdb88d12356","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-29 09:28:03.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2e7f3dcccedf86f85cdae7e31844989e862e09ed","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-29 12:12:48.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/206b057ca0cd4abf95e7aaf1f033cdd5\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/5bedaa3348234a27912228d02b4cb4da : SUCCESS in 2h 04m 48s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d72a3db457314a0db700875e3a44dbef : SUCCESS in 2h 11m 29s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d2c46d3daf4c4f82a77918901e4d08ee : FAILURE in 2h 43m 55s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"7cfea8e04f9177d6f66e14e2bb124c6927663b44","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-06-03 17:02:31.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"aaee114aa79cf782d56a9a5a6d4cfef5c8097128","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-03 18:59:18.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/0a58aaac59fd45fb8e495f8a216241d5\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/08f111f80f464221ada92023d71d07df : SUCCESS in 1h 47m 26s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/5bb55f192b88464db370923d62a934be : SUCCESS in 1h 55m 51s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/02fe878c433d4e1f8f6bf816608546a6 : SUCCESS in 1h 45m 26s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"7dda01a41ebdf2b5dfcb7cba00d3515920ed7322","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-06-11 07:45:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ea287e3083dcf2bcefc3a60e88f38a3f01740fd1","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-06-11 07:50:50.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"803b26b6e8f1762e524856b64c622c5f5cd62850","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-11 19:29:00.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/0725076f0c584edabd399a60a234e69e\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/47941634ac3144838aadb45450daeb1e : SUCCESS in 1h 50m 16s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/15170202c73046c4b5d60be4f9930c75 : SUCCESS in 1h 52m 05s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/fc6397b3733447e28cdf28c06d9f15f3 : FAILURE in 11h 37m 22s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"4d2f0afb447ad6238ebbb31fa845f4b9fbec9794","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-06-23 13:23:45.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2e462a3855d711d6547146ad0ca856d237f524cc","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-06-24 01:45:56.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/077ded47fb9147018e2b2c75ca36f6c4\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/3fbe9b4247da43fb9ac640f2f136ba9d : SUCCESS in 1h 37m 53s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/802f3f76d93d45e893a5915ff39c2fa1 : SUCCESS in 2h 04m 31s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/a6bc43c2890f4aa6984475786aa9f8d1 : FAILURE in 12h 21m 21s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"790723f869de6bc1a2c8d7f32ebc8fadbd0c6b26","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-01 11:18:39.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2c64dce541c49cdb1772779e98dec9b556146790","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-01 14:47:06.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/981ed1013077474ebcdf5cb2000df682\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/7a61e6668d04493fb31d402f761db40d : SUCCESS in 1h 48m 15s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0445d41628b646cf9e0c780e02dca2c4 : SUCCESS in 2h 09m 23s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/1ddd030bd4d7422396095978ab0ba3de : FAILURE in 3h 27m 30s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"6e3eea0095775547269ceb6c1bb57f59fefa117b","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-07-08 11:24:30.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bd06dbe6a158ba6e7143f1be64acf4c31aa11b23","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-07-08 11:31:41.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ebf2647723711147da734e0c13e6984de22aff9b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-08 14:15:01.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/612de289ac2b44f590657b30cdad2c25\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/07bfb68d4f9c44f6b3205e429a7714b6 : FAILURE in 2h 31m 39s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/4006a36fcc064fdeb0133ab221623b8c : SUCCESS in 1h 56m 10s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/7c6dd66936d440e4abb9f84c075010e1 : FAILURE in 2h 42m 29s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"0f9645cc994f30a90009b9d494441b252f1e223b","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-07-09 11:41:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"460947f5b519c3b745b0528e904dd56733656c09","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-09 14:02:18.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/5176d9f6feaf40448c6726873ea8c28f\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/5cfefbb3b5a64499ba534d7a84eefbc0 : SUCCESS in 1h 36m 52s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d98a95bc7a564dad8e9de759e14d0c4f : FAILURE in 19m 47s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/9d89186c197e4e238da1538e1a3ed3b9 : SUCCESS in 2h 20m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"688f77ad36e7cff92e997487f1c2bb71426c2b85","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-07-18 04:59:37.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5ef6cf301404a88aef1756d7233c3485b1239587","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-18 07:18:25.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/d51c33a27ec84b24ae2433680d9eeb2c\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/1ffcfe9b4ac543ab86eef57ad241aa11 : SUCCESS in 1h 34m 57s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/b6cbdd9a82634c41860e8febd90d86c1 : SUCCESS in 2h 05m 07s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/529338b95ba04991a89e2a6e03a5567f : SUCCESS in 2h 17m 33s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"8e54812d7b93e77e6bdf39a1d3c5d85005fc15c4","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-08-25 10:02:48.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b121fe812e6e136089e39d935d2bdd6e6753172c","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-25 10:19:23.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"29b4d2f3f1f2ce10b5ad37ded3489cec5b232740","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-26 21:29:32.000000000","message":"Patch Set 15: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":15},{"id":"189dc6bb38768bb5fd6e1c4d3c5271d0a8515584","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-30 05:30:10.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"abb842cfc83abc270424d923d5a46bd9e4ce596f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 05:57:27.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae4490452eb946ff841c0e5ac7ea73fb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/61784698f90f4c54a7421be06966fa4f : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6087d750a5714013929f5bede18df5ce : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd49ed3fd7f3483fa9101db4cc299e48 : SUCCESS in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"88fb89a3ef6d2e66f60a1669d990a1fe2b9fb830","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-30 06:25:26.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/22d4d605ff9d4870b0e27a771c8573b1 : SUCCESS in 52m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2aa75ce852c1491196b5f193776594ee : FAILURE in 50m 18s","accounts_in_message":[],"_revision_number":16},{"id":"ae290e8089a14e6e55f9a2e2f5df748d18e6e1b1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-30 06:27:07.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":16},{"id":"edb44e9b834d1d49b70d92a8f1985e7c7c290a1c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-30 06:49:46.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 10m 00s","accounts_in_message":[],"_revision_number":16},{"id":"760e35907890d3fb3f60ed2cc67cda00218e60e0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-30 07:18:26.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":16},{"id":"4eadd0bd9efce141f1748151da47c22abb65a74f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 07:23:25.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36a74c5081a44b1a86b4e52810af6a81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e923a8c3b54014a86017670d313510 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eebf7b536ebe4a1bb055844846846cc7 : SUCCESS in 9m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c1eb282315744da826c14bd9b24e5bb : SUCCESS in 10m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d3ea0407d114fb29c878ad00efa29e4 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c96c9d011148a4bc89479a32ca9f97 : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8705254769af463095250ba5f26cc3a6 : SUCCESS in 51m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b3e84b51856b45649d0ade51e47be0d0 : SUCCESS in 44m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90093fa8688a4bf39367794e23d89abd : SUCCESS in 1h 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c121407dca55449fb053ca5671bc0620 : SUCCESS in 48m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b87709c495444d3a95f609a1f07fbae : SUCCESS in 11m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f5d3442615684929bedff7eedfad69ca : SUCCESS in 23m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/55dc4821c85f4305a3ea529261ee5995 : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e157589f6b947679e41291a31ce6782 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10e1a5d977f143ada4caec14d589f4c2 : SUCCESS in 5m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5d3c650d6f374b36a154ca83b6e4b9e0 : SUCCESS in 8m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1d13ce5f2594ba198a969d4fd62b0d7 : FAILURE in 37m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/70a10362c3a2469aa7f97f26a7366210 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9ce7a716cd2d40179e5bbc063bf1103c : SUCCESS in 1h 14m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/061b0e1ac36e48f0afe70ef67f78e749 : FAILURE in 1h 25m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a993c7c0c2224c82b87a210898e352e5 : SUCCESS in 1h 48m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9393b1bd641c43a1ba3ffd4638e5a6a8 : FAILURE in 9m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2c91dfad7479471c8fdc8268c379e2fe : SUCCESS in 24m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7961b5c0b6ca4273a46bf3f520050cf8 : SUCCESS in 52m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8630d4f7922b4304801331fad0775755 : SUCCESS in 1h 27m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fc5fb8d123ed40afa1306165dd534cdf : SUCCESS in 39m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/30c29f1b97f44317a0718b658a6f3a6d : SUCCESS in 1h 27m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f13ceab0839e4352b15a0f5d8329747b : FAILURE in 8m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/05410602c8684ae9b026a65b2ad8c99b : SUCCESS in 1h 24m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f3daebcc2034e818ccef926dd78b290 : SUCCESS in 30m 57s","accounts_in_message":[],"_revision_number":16},{"id":"7616bc050821d0882ade43690a24cc5f064e81b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-30 07:44:32.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"edf384eea1edfd0e89c8435203f1073005bc7d7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 08:06:59.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7db00f72e2a486eb5d876372bf07818\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e5f54825d414358b3ab5cd4a29b02d7 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aba7c4aa637e4988aa64879fe08d34b7 : SUCCESS in 18m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a3cfd307c4774d558e60265fab901f0a : SUCCESS in 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"14fed80cc62057f96f4ed92e1aaae1cb8f022038","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-30 08:42:00.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912058 : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":17},{"id":"d4b32e327c23b618c083fc92b1ec7a9afa69c9ec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-30 08:55:27.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/35428e69b1804626b58ac356e3b88062 : SUCCESS in 48m 02s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/655fb02cebf24f7a8bba7ee93fc7cbbe : FAILURE in 1h 07m 34s","accounts_in_message":[],"_revision_number":17},{"id":"7757a8aceacd6bd4413ea185d585c97698a76c8f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-30 08:58:21.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 12m 20s","accounts_in_message":[],"_revision_number":17},{"id":"aec836f2f55f4cc67a46a788541786daeb93bde2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-30 09:34:35.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":17},{"id":"52ed00131589709a5461ea4c0a811a085d9732df","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-30 09:42:52.000000000","message":"Patch Set 17:\n\n(8 comments)","accounts_in_message":[],"_revision_number":17},{"id":"3fa0069556317f022999c5cba1b5febd243dfd7c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-09-30 10:53:42.000000000","message":"Patch Set 17:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-17/FC/ : SUCCESS in 3h 8m 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-17/iSCSI/ : SUCCESS in 3h 6m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":17},{"id":"a4da5cc5f1917febcc32de29b60fd0dab7b188d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 11:24:17.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/594279e041b24504a8d96f519d1048b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c400482c3c4bf48570fc4a793d95a2 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ea7739540f04c4c9ff7500dd58c99ec : SUCCESS in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bbfb62532eea4b5d9366953e09dd1512 : SUCCESS in 9m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc698a625c7a49a6925de8f2927e84a5 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2640b16fa64aa181454477169036a1 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/c4c22616f7414bb48d7ec2e9df5abbc3 : SUCCESS in 46m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af5aae90cabb4cfbaafd7b773eed0acc : SUCCESS in 40m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1d16078df444266a63141b50a205444 : SUCCESS in 1h 18m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c29f9eb96c34f769cd19b1245e68a76 : SUCCESS in 1h 03m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efb0e0da72594e1dbb3f8212f2052adf : SUCCESS in 13m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/098bb45ca2da47258648f0dfd4f62b52 : SUCCESS in 16m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5c23ac4ebf9f4cd784a8aeb2ded7ebee : SUCCESS in 4m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/024fdca226cb42468885a4d9d8eaabc0 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07c22b8d8a0e4af3a77acb46fdaa6da1 : SUCCESS in 5m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/760f0f4aaba64d25b98d6c6b4c87df73 : SUCCESS in 9m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c69b74866b784537848b847186a406ae : FAILURE in 35m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/344b5db34b194f04aa9dd0ec6287bc4d : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2557e18131b44a738c2f4f7fd5d36acb : SUCCESS in 2h 07m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a7800b02d9274263ae4dc70011d3eef3 : FAILURE in 2h 45m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b946749216bd4c979dc441c40b5dc79c : SUCCESS in 2h 14m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/002184b2ab8d47e49a94d26e6a4c146d : FAILURE in 14m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/024640e97da048d2b3a3a6f96ca17786 : SUCCESS in 34m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e217ffecd6a438098044682808daa40 : SUCCESS in 44m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ebd4fb258bde4950887d95eee53e4e7a : SUCCESS in 1h 23m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1f9b1427155647469681e4553eac4ffc : SUCCESS in 40m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0d857b3ff27e47c9aea3b59339eaf850 : TIMED_OUT in 3h 34m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7cee72f251bd4c5498ae6ed781caa94b : FAILURE in 19m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a09858c208354b2e9bd5ff2a9540dd29 : SUCCESS in 1h 35m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/592675bad9444634b4f4f3abe36f12dd : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":17},{"id":"a5f2fdcbace7ef84b7d12a4f3343ef8988dfa609","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-30 12:44:31.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"ec8cc16d8bd08329358cf30eaf9697405fe111b9","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-30 12:45:11.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"32d8566493c1a0766084d873c90075fcd389af97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 13:10:44.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8187fd5a77f44d18c8be9cdfde06790\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c4dc4607b3a458089ffbbb974f32808 : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/26bdad4985494e7cade4674928ae4e5c : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d679726728f45bbbf76cf759dab6a70 : SUCCESS in 23m 46s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6c7425b60298dbedb9532c6876af5c5e15831324","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-30 13:49:45.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4a1450985efa4f8a8769fdad8423426f : SUCCESS in 45m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/96f34f7d9ef844d8ac4df5bcbaff1de2 : FAILURE in 1h 00m 56s","accounts_in_message":[],"_revision_number":17},{"id":"87a814c48c0e92b1225d5e288866883b466ccc20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 15:30:39.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c51eb0c78ebf4af0b0869eab02ca98fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a34afd8073c4ce3ac81e7e392d1db81 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e36714f4ad6d44a7bc8d85177a6eb244 : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55408d592a5440809d9bb2b79bc90841 : SUCCESS in 6m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aaa4208cda5048d5b708d2891c16abe4 : SUCCESS in 14m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee02d981543e4413b33810d9aae790a1 : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/17540a0eb5314034b93e05be907727a2 : SUCCESS in 1h 01m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed32944cac8344a7a098b5b319bb6899 : SUCCESS in 42m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/034d77444e3243cbbcb01d6eb7a6d5cb : SUCCESS in 1h 28m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34f979e8e7d947869ce1fb72da8a893b : SUCCESS in 1h 03m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5221d32d94254cef953877e2c7d4e64d : SUCCESS in 13m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e2111873b1e04a62a1d77d1393d1ec78 : SUCCESS in 22m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd1643d44e54413e8e1bf9124b2513b3 : SUCCESS in 3m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4ef93cebb1834b228d070466e8ba7285 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62a3f1bbe0bd43bb87c2695939105fa5 : SUCCESS in 6m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/715d6337966146bc90e80dd8058f66fb : SUCCESS in 9m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a77db4de34164d00bb8624262eb51c68 : FAILURE in 22m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db3440fbc69447d3ad4ecb2a00e92737 : SUCCESS in 4m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/824848b8f5644f5fb0c4d6e00f47a06c : SUCCESS in 1h 56m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/08ba52ad7964455b9db68a10324c3e71 : FAILURE in 2h 33m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/45bddbbfc48e4cb4a9cfae10d43e4a26 : SUCCESS in 2h 12m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0457f94105874e3a9b08205603f9d427 : FAILURE in 8m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/132e6560322046978e3f046e9b9f4f2c : SUCCESS in 16m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd5a553da41b413f8f13c9ac51c551ef : SUCCESS in 1h 08m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/51ecac3dc131477f9cfd71397fc98171 : SUCCESS in 1h 32m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/88c26174a7064aaeabac6b1042741037 : SUCCESS in 26m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6660b43c1f294f029529f75abc45b6a6 : SUCCESS in 1h 43m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0d3723b05d204d0cbba02444ca41ca4d : FAILURE in 15m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/788d1facd29740faa8f1728ba172003e : SUCCESS in 1h 38m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3b8798a3a3344dba34f96427f3fd262 : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":17},{"id":"ed40f8fdd4322463664895d16851c4082057e53c","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-10-09 11:17:00.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5d0c1f1598d33d314fffaecf26a166cda2f0d1b6","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-10-09 11:20:01.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"237d3fcb533ff511659f5b071c9c4da2f163f16a","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-10-09 11:31:02.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6dfffbbb3b8e7427c1bcfda1f1442a772d1e7c64","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-10-09 12:07:47.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"27701a7325e109cc457053c0e82e218017b0d49a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-09 13:51:01.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"576a1a77531fee1b4d3a1782f3618fdc91dc3c4b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-09 13:51:24.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/8eee232b02ca4e9a8fcec8c8b2e6ca4f\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/4535808db3b5400eb80ab08d2f1a4c66 : NODE_FAILURE Node(set) request 200-0000000000 failed in 0s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/9d16416e8f7942729f6f0e6dc0a7ede0 : NODE_FAILURE Node(set) request 200-0000000001 failed in 0s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/36b0ee00d04d420abb2ee6bea12d3a60 : NODE_FAILURE Node(set) request 200-0000000002 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"e8b43fd39294605ea88a2cdacd2713c177d277b9","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-10 07:01:52.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"dbfcdc97d14092a2da4dd3e7f1b71047d3aa6e22","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-10 08:28:56.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/eb0c31ce34c14f82ac01e1dc362b9f1a\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/7f4109068046490ea65feeb098712dbf : FAILURE in 1h 26m 40s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/bbb2cc6444454c19ad304018fca3c476 : FAILURE in 1h 23m 30s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/b57e3987e43c4d8cb966d3b6d8e1c26a : FAILURE in 1h 22m 29s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"3acf8d8e14cb57ddd1f1cad236f20e2049e629d6","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-13 06:18:39.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"31ec368f8bacba06eaecf4bae1524b1996d30a00","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-13 07:44:26.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/9f27610f61d046bbbcbbb4ead5062e89\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/323fe51b92484b8088148eb1bcddfd71 : FAILURE in 1h 25m 21s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/a142d980e9f941c78e5d828dac273f01 : FAILURE in 1h 23m 32s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/1ad9e9ec4b504f7284b5edb1041b98ef : FAILURE in 1h 22m 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"375a6039daaf63ead249caa76286cd0aad439c9b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-13 09:26:04.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"772287f228faa75f90e2abaa2cccfb038213bc63","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-13 11:06:48.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/57aed321a235437ba95da80a196c400a\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/90c90c6c87994bce94693b689fb4def0 : FAILURE in 1h 40m 18s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/e2e1caff61044b1d8328a44a3e84dc37 : FAILURE in 1h 39m 48s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/6c1a19de00754017be8fe7320d908b70 : FAILURE in 1h 39m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"6fb11fe6ea42c12811db3945d3ed73f5002b43cd","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-13 20:44:21.000000000","message":"Patch Set 17: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":17},{"id":"7db0f5cbde475e2b7f64bc8721866884c99dde57","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-14 08:15:22.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"0ccf2e37f4dfa2a465a4a7946c103f868da44921","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-14 09:42:12.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/50f2d8ceff4f472e823c3746dd18da81\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/fdcf0170e7c94ec69505add833eb47b8 : FAILURE in 1h 26m 27s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/1e868c776de14f60b7906e030af807cc : FAILURE in 1h 23m 59s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/82974ab876584950a45d3150301b7b51 : FAILURE in 1h 22m 21s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"6f2c26e9aca7455b48f3bc4d67a9b2eb0e0af960","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-14 13:29:41.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5a037816f006ce38cb56e7efe15f3538027a1ebf","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-14 19:15:59.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"11b9947e13ba996b8fb5789dd159b79753c518a7","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-10-14 20:05:04.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"fc7ced3f18e7b40ae32ce3c3a63d7a08ef2fc24f","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-15 04:05:51.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"edf5457364bd0cbc459192150ed1c68fa54be753","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-15 06:14:08.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"9e6c9115634297eb12348187fd6454274d81d127","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-15 16:21:03.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/4259b34933d74f9ea8bbd084536fc4bb\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/36a38961aec44197a2c361feb8935450 : FAILURE in 1h 22m 09s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/c7f60310caa848aa957f159491d79751 : FAILURE in 10h 05m 55s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/2346ce50912b492e83442dd2e6d5baa5 : FAILURE in 1h 25m 14s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"281aa95396ea2e601e4e91cef65b2f564dd5f9db","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-15 18:46:51.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e7c85339529444ab68d0824bc91cc84126d7d8ef","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-16 14:15:04.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7d71640d0f4a5a5bf63bd8f52ef4df5b1aa748ca","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-17 06:47:31.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d6fc24c5e6ea447d4abf4765e2fba5c5a5acc5c6","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-17 16:46:19.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/a4777a53a6b8480ea626f6e69dc887d4\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/b5e5a93e2763474bbe575d36d76d1472 : FAILURE in 1h 37m 36s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/604768515a0845feb6423785ef1968c3 : FAILURE in 9h 58m 22s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/83e9a66943504fa887d232ea31318892 : FAILURE in 1h 37m 08s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"c5fd134d1c6c620fb2554210059d8d762543f82e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-20 06:25:47.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"a5cf0c6a330d3fb742a26a9ee051f87b2d7df2d2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-20 06:26:36.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":18},{"id":"28b30620ddfa509158e34bb543d401b430b37f7d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-20 06:27:41.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fedcc5f8ba704d14be7225e57d67b0ac : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1b3c6f5c05e44bb8ba2c31f396031c7e : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":18},{"id":"ef942218133fa4c510ba67c998706d3b668f2660","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-20 06:39:15.000000000","message":"Patch Set 18:\n\n(9 comments)","accounts_in_message":[],"_revision_number":18},{"id":"b692ddb09f363fd3ea9fa420c4ab0230bc66d4e5","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-20 06:41:52.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"6eb612ebbce55069d3fa8fd1e61bd680fb575d69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 06:51:08.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9288a68ba9864b18b1b114d6456623c3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e39ab39e7b534b2e98e9d0e169b8dd49 : SUCCESS in 19m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e95f6aaa7e144f368119d52dc8c679c3 : SUCCESS in 20m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/18f8423e03f64efeb9fd2df37f82cc97 : SUCCESS in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"2cffbd77f95bbac8d4ba0a77efdff79dc218961d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-10-20 07:03:58.000000000","message":"Patch Set 18:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-18-fc: FAILURE\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-18-iscsi: NOT_RUN\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":18},{"id":"c8ceeab231637bb39e8874b308c72757dfc50adb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-20 07:24:49.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":18},{"id":"d3434bba8843d17f101c9d00f3c65a502e70f005","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-20 07:26:42.000000000","message":"Patch Set 18:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2079/.","accounts_in_message":[],"_revision_number":18},{"id":"b1828458572da0acd412110d8e79fe5911716524","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-20 07:27:43.000000000","message":"Patch Set 18:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2193/.","accounts_in_message":[],"_revision_number":18},{"id":"760a1ca782621a76587da12ffc140ba467b72bb6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-20 07:38:03.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 10m 57s","accounts_in_message":[],"_revision_number":18},{"id":"2ef8ec81b6534eee1b1ff7b099f7b973d82686c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 10:04:55.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/085ebf48807d414aa6931fec8566076c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2fead64e5ba4713a16d0904231bcfa6 : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b574224f96244d099176af27d09372b : SUCCESS in 9m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcd8d070ecda4cc4b833e8469af439d8 : TIMED_OUT in 40m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597529129d7a45b6a585444cb7d4b4a8 : SUCCESS in 10m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/be882a3be8a54c0aaa1a6aa2f34cfdc6 : SUCCESS in 1h 02m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35f5852923724b3589d2d275edd0196b : SUCCESS in 41m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f89d69191dd4c3aa885736210bbd6d6 : SUCCESS in 1h 29m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13873a04bfc045ae936791a87a246a62 : SUCCESS in 1h 03m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f85fc3cd0c84202a16f5a1fe18e67b9 : SUCCESS in 9m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/887f37cc9b5943fd898e7635bfcaa0f3 : SUCCESS in 19m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab4d678c8ad4442bb6f88848e0840b22 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e6754f8f178240a7b60e58f7b3b68568 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3fa244e065bc452694cdc7c4ff552fbf : SUCCESS in 9m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/02a3f834224e4297906cdd5b7b5115dc : SUCCESS in 7m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/41d25115d18f425c97950c19cfdd4e03 : FAILURE in 36m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e92c2ce9e429400da557e11f4e6d420e : SUCCESS in 4m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6ddcb45a758948ff982cbe77a3447bea : SUCCESS in 1h 24m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4399900cf12d4d91839d24739ff94b00 : FAILURE in 2h 47m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2bc717f82bff4ad8b68888efb7a08930 : SUCCESS in 1h 57m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d39cc4764817483988f56808be2ef0a2 : FAILURE in 1h 53m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/64b141d4b8694cd3a810f2189e30491a : SUCCESS in 32m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f24e69b010fb4b538ad4fd64d187fcb8 : SUCCESS in 1h 07m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fad824b23bb44b599b10f5b361680d54 : SUCCESS in 1h 38m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86a2d045a225486191e0a79e0c1a9d86 : SUCCESS in 42m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1e1d0de5729941d4aff8a38fc94ea44f : TIMED_OUT in 3h 34m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d4ee84b06c6e4547b32a1739c38070b4 : POST_FAILURE in 2h 12m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d955d4aa761d444684e51fbf450ee704 : SUCCESS in 1h 34m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d3c9d4f5950489c8ec21f32fde75a8a : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":18},{"id":"7149a458d9a6b483183b20e984db60ba8fd7830d","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-20 10:08:14.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a7473d762ba1dc8dda884117735e40c28f72b705","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 10:32:15.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c56843bc7404e6f98e3bb07d7d17071\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8013369737d146b689a415a596c98d1e : SUCCESS in 19m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e63e78b02bc449d96d721f1a5207955 : SUCCESS in 20m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f02a6a4c7b944aa4803e617518fc5eac : SUCCESS in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"f3cf2a846712c079f5fa377a88b2babd99506a34","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-20 11:11:12.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/39c4f6be7ab444ce8d0f4b80686b7e40 : SUCCESS in 45m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3f9283a62ea04dc6b4f6d26471b396b8 : FAILURE in 59m 34s","accounts_in_message":[],"_revision_number":18},{"id":"4e3195857abc227ba45b3d22279aa8c54922ab8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 12:44:19.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5da3cd9e4a94a148633853f50554145\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f685e24ef6584dc0a79160ec49b08d52 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b98c811f5f94405a984350f7192dd1b : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bcb3a2813f940099777a83f68003553 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ccc9bc67cf34a10972ce5cc6044dcbb : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/4c1af0df7ede48cf8b5a39d2c02c1e3e : SUCCESS in 39m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ae66fed7f9b4ca6ae260fb1787ca466 : SUCCESS in 39m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c90569261e5e46efa9847e79c171e877 : SUCCESS in 1h 24m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3dd569c3b86465cb1447e29cc66ef52 : SUCCESS in 1h 00m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50028ada25ba499fa509a00757b04ae3 : SUCCESS in 9m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fb687af98a7f436b91f726b0fe67b1c7 : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fa134d25ee194962a16985e2ed1d11a1 : SUCCESS in 3m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c72dd3079034493977b5d2f2a7f6548 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/59b2d4af004b45308f3ff121bbc57745 : SUCCESS in 7m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6b9b513f7bcd44e7a6eccd972abed040 : SUCCESS in 8m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb74c5a68a764e818cf19c827742ec7c : FAILURE in 22m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ae4557f18454e08bd984739512b7968 : SUCCESS in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6e9426de8eb74ddea30519155712ad27 : SUCCESS in 1h 53m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0b7c28e2cfc84e038bcfac23c3d57815 : FAILURE in 2h 26m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/23047e4be73b4c80a780352d227ccb66 : SUCCESS in 1h 49m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/014520acabc9485588689448d9f3a6f2 : FAILURE in 1h 27m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b9dd31890a304001a206028d9e37e401 : SUCCESS in 35m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/605d50debe514b56bbab7e0a6a94e79c : SUCCESS in 41m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4f50dd6f40bc427db13e8c64412f1cc4 : SUCCESS in 1h 28m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9f16c170b4f44e358efa57fb966cd4ef : SUCCESS in 39m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9d921fcf5e43428092cc989fb95de06f : SUCCESS in 45m 57s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/80cb601eac9949e3b12671d22c1dde2a : POST_FAILURE in 1h 35m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6808aadfa30843b28611756232cceab4 : SUCCESS in 1h 51m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da20031781b84bdf910b9d2e0715d780 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":18},{"id":"766e859d37e2084f2c0c556bd209b3255ea6717e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-20 12:51:02.000000000","message":"Patch Set 18:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_10_2025_12_44_28\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":18},{"id":"a0e7ba9de897ae5947b8f0cd23e67a6c3f8e1119","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-21 08:26:03.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"0fc4b0db9d0843b6d686fec5d9e4184ae6ef08ce","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-21 17:48:37.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/a24acea6a967444e8c9bdf55d6f3fcc0\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/730b04a9b4ad4db99c8eb5355aa36c06 : FAILURE in 1h 38m 37s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0db478fab99149e5a7c774d4ed7ae1d9 : FAILURE in 9h 21m 46s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/2e42813ed3454012b1d2278b74bd6df6 : FAILURE in 1h 38m 08s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"a030e62ed9a2b5f60e890f7371c1d6cff9288ba0","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-22 20:06:48.000000000","message":"Patch Set 18: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"b25dad83f810dc8b9ee5f6685811170ae84807a1","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-22 20:31:55.000000000","message":"Patch Set 18: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"da566270eeb896a94f3fe208f9fc2427d85a135d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-24 06:32:54.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"bf6379939c91688116035dff8fad818a78305abb","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-24 07:35:15.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"2510cdc8dac4f829392d4c5ed7b1427fee46da8e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-24 07:57:40.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/0aee06ca88c84335bd729adcb568ffd4\n\n- powermax-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/3c3001a0cf92475385f39481d207023d : FAILURE in 22m 01s\n- powermax-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/f76dd47200ee4b7e953de20b3d017b89 : SKIPPED Skipped due to failed job powermax-iscsi-cinder-tempest\n- powermax-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/9a80884c76284c91a64789b408822b39 : SKIPPED Skipped due to failed job powermax-iscsi-cinder-tempest\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":18},{"id":"45f104589511aa25e79e80d660dca804619ba9f4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-24 09:18:59.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"4e87b4a1dbbf47d95c7e5de4fa73f0a1e10463f2","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-24 09:38:07.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/8ab3e3010cf240f88a609574bf5ca05d\n\n- powermax-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/aab55dd0e3264684aa8c171e877505c8 : POST_FAILURE in 18m 46s\n- powermax-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/a390634007a6460da84042f09bc361e4 : SKIPPED Skipped due to failed job powermax-iscsi-cinder-tempest\n- powermax-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0cf49037e99e4205ac7e6260ac225de1 : SKIPPED Skipped due to failed job powermax-iscsi-cinder-tempest\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":18},{"id":"0fb721cd32a87b4d237cad469feb70fd335bbf8a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-24 16:14:14.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/70a39cdf3c37476f9754b23c0f516865\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/51d5e3c2770e41dcb704a7af3ebadd46 : POST_FAILURE in 1h 23m 37s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/7d9674235ef5494ea219521e6bb2bc39 : POST_FAILURE in 9h 40m 31s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0438e310dbdf45efaa7e448b8995079f : POST_FAILURE in 1h 24m 55s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"d718424efc24125ebc7ea8f73aeda5cd7a7da3de","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-27 12:21:18.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"8f14880c555af5a051cd77ea3acf410a8a917764","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-27 12:54:07.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/826bff00d31d47b8abedaff228775182\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/e4058a8169e14fc69e6f5caae93bedae : NODE_FAILURE Node(set) request 200-0000000074 failed in 0s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/6a6285b0e0354359b7285fdcc72ec107 : NODE_FAILURE Node(set) request 200-0000000075 failed in 0s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/006e43f2281e4411beb54bd0f55d986b : NODE_FAILURE Node(set) request 200-0000000067 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"65cacb351ef26b79922f5e15460dd2c703a49f45","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-29 09:08:16.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5ed3c7ab4580f8b24c56bff3d7572e39beaed826","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-10-29 10:18:48.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/e1143c46c9d04f738e01a40b85240350\n\n- powerflex-v4-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/ba51316465d44b328a33ef8ba3225ac1 : NODE_FAILURE Node(set) request 200-0000006006 failed in 0s\n- powerflex-v3-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/69bcb9d464de47cd8fa3b03da8d61beb : NODE_FAILURE Node(set) request 200-0000005995 failed in 0s\n- powerflex-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/ee3103623fd840f8898df6aee2f7c8a6 : NODE_FAILURE Node(set) request 200-0000005996 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"167a22f83b68145ae85df59fab67c624b73e3b74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-01-08 08:32:16.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"ca9064664f75f0da29c6b4bb5f1737b9e83fc4da","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-08 08:37:53.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/595b047eaa714ab89073de5b5f989e53 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cb4fa5706b1c461a8d75337082589a45 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":19},{"id":"03740a72fdf7b8934a4043fb9a1116bf9b5bfe73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-08 08:53:57.000000000","message":"Patch Set 19:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/142/ : ABORTED","accounts_in_message":[],"_revision_number":19},{"id":"553ddf5e7fc59697d267a2e5a05ab99e02dd4d25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 08:56:13.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1b2c3866b35464a80e6d743c0283274\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30c17b1b4e054368b2c2078428758e0d : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ed6b81ea6f1479fb546d20bf009f646 : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2bf32657f60b44f49b3452901b8aaecc : SUCCESS in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"2d51a7e1e5825edc50902bcfedc26578196c35a5","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-08 09:10:45.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/9193114b6b6040a8abb69ebd47b665ef : FAILURE in 38m 00s","accounts_in_message":[],"_revision_number":19},{"id":"748a55936e8496d31c0f199166f6e6aed6ede9a5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-08 09:27:08.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":19},{"id":"f64ac44c63f40e263a1b769bd8af70c4a4c38a4e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-08 09:53:01.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 19m 32s","accounts_in_message":[],"_revision_number":19},{"id":"1bb18c38b3bc9c3499ab23450bec899591205613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 11:49:34.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18f1241763d142f0ade0d179f1ef78f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42246ea2da7f48d08250fadfcf0ce438 : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb636ee1f3ad4bd39957a5495aa5c42a : SUCCESS in 7m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c0bdac2d8ea486a8782d52ebcc72e83 : SUCCESS in 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21e0312a6744a0abb8fe3d3b045da24 : SUCCESS in 14m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/2eebbf9db9fe42ca8611777453fa3e2d : SUCCESS in 59m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a85c39ac3d064fad99eba003bdb3e2d5 : SUCCESS in 1h 00m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f207d92e0fa441b9bc7c9eecec87bcea : SUCCESS in 1h 35m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3674e1a2b12e4792bd7e3586014a561f : SUCCESS in 59m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2d361b01f534be88bfe62759d8bfbfb : SUCCESS in 15m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0a148b0aa3664a5f841a8e4ac67978e0 : TIMED_OUT in 40m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/17c87e5a4ef54e3aa474c62ad8a530f5 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40272237ac9c42b394cab2ffa97b41ca : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a21b30a8e06f471991df7b9e99ca361e : SUCCESS in 5m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/01c36b33cfda485682985b3c71eaf8d7 : SUCCESS in 7m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bcab04c6da4847b3adefcb5aa2149fef : FAILURE in 35m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d62084f7f32c4b72a621ef776360f7bc : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/36849f79fcba417180106a072c167b46 : SUCCESS in 1h 48m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d4485c3b8afe47b3a7db1db3031b509b : TIMED_OUT in 3h 04m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/17ba44f87dbc4273bb2b0250fe09f868 : SUCCESS in 1h 55m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/aeaae4668edb491ebe6a46b731d3eeeb : SUCCESS in 1h 59m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ad60c3bb359f415e94ea131afe60e7e0 : SUCCESS in 29m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce0bb0495f09484c9a60c6149efce3a1 : SUCCESS in 44m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d391a4210f354ffa995239bd603e714e : SUCCESS in 1h 12m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dd1e5aa4aea84c0a9f0e9ae7babf0180 : SUCCESS in 43m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3d78252b1e5b4a1d8b78f72044dc06a0 : SUCCESS in 1h 56m 50s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5284f44162a541adb3f563c88f3ad0a4 : SUCCESS in 2h 16m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b32861557d3412a8312df33b9517865 : SUCCESS in 54m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eae2cbad1c654d61a408060ee6de9f02 : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":19},{"id":"cb1832f8fb9d5a0d7f20d741265da2ab3b488d93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-08 11:56:30.000000000","message":"Patch Set 19:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/08_01_2026_11_49_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":19},{"id":"b843f147fa047dcc6f6b2be8a8153181fe006efb","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-09 01:46:39.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-19-fc : SUCCESS in 59m 48s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-19-iscsi : FAILURE in 1h 21m 24s","accounts_in_message":[],"_revision_number":19},{"id":"0bdf75aef34144829978032b5c13b71a4aa56409","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-01-09 06:37:17.000000000","message":"Patch Set 20: Published edit on patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"8c158bafb4916c8e2ef10b932fdd8a81eaf5c646","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 07:17:33.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909aee4f47604e6ab886818d64af7d08\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0de239319ef4290a5eb6ed547f8937f : SUCCESS in 14m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3973e7c9910c421b98babb6ac6792f50 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84c53f09b7fe47bda280294e1600d0a1 : SUCCESS in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"9afbdff44aeb49d5ede14f53cbc12ed396b2a884","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-09 07:34:52.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":20},{"id":"fa20043b154335c5246e3e12368762d579ac886d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 08:01:00.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0fb97c584624c148b361da4bf7ab2d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37206a17063445939def04c8a5eca14a : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37344c494a6d4882938f7e52d4538f99 : SUCCESS in 7m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f56eb10bddac47d998d544fd4d89f855 : SUCCESS in 14m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad90470ac46143d1a68efe0be8214f74 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/f2ceb5e149ea42b385926b623dec44f4 : SUCCESS in 56m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f0fe61e96cc4262b7be64f88cb36b8d : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b82d37b3c564d6f8b586317306da7ab : RETRY_LIMIT in 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2b086cc2195497286d0754926426631 : RETRY_LIMIT in 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b88d074bfda244d494030c9c836a9f1c : SUCCESS in 16m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f6a98d33060a43dd9794ad047cf2c6dd : SUCCESS in 16m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/48eac775e3f64d3f9ffad1b28fe22f7d : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6cab36c15f794d0d8e8d56da4e942067 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/72b0414c8820452ca1e842cf5c847be9 : SUCCESS in 5m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7d40e971af014f79b2218f6cc4593e73 : SUCCESS in 8m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6004a56ce0a845779fdafbade725b8ad : RETRY_LIMIT in 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f9be27c75f74827bd377e69ecc8c138 : SUCCESS in 3m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8d60f2a934704064864f12abd4d3a71a : RETRY_LIMIT in 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4deb4304f50f43c288cea21ea1ccd8fa : RETRY_LIMIT in 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/50dec81e66924c38b9dfbc9b133d204d : RETRY_LIMIT in 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7fbc89820ab44f73964e4ccd76c109fa : RETRY_LIMIT in 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5a98cfaccbc84a3bb26ad9bd4b2e3360 : RETRY_LIMIT in 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bec72d6b213940c49c231713f800f154 : SUCCESS in 1h 15m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e56367e668a945a4b590c703146e7fa1 : RETRY_LIMIT in 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8f44dc5ac32540658aabe5282a7b0cae : RETRY_LIMIT in 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2467454b16be49a8ab4adb210c9784c8 : RETRY_LIMIT in 37s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/19de261311854d3fb20d21ac613a00a6 : RETRY_LIMIT in 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/497cd403f23141c2ad4b340c15eec45d : RETRY_LIMIT in 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee136e3aead14713aeb639f8ee5d77f0 : RETRY_LIMIT in 38s","accounts_in_message":[],"_revision_number":20},{"id":"d8a3a8dca7a438ddd3b41027e8ba40cfaef71e82","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-09 09:02:05.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/80e8f441f21947068d747bc059b80cdf : FAILURE in 38m 05s","accounts_in_message":[],"_revision_number":20},{"id":"f54eeef489ccd1573d6e9c5f37cd7a3e4139940c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-09 09:49:52.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1ba44ac509c34931abda6377cea2dd41 : FAILURE in 56m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1d63cb4924574bae84e284692e9c95c1 : RETRY_LIMIT in 10m 19s","accounts_in_message":[],"_revision_number":20},{"id":"8d0eea97b5c16cbc3d1085c1cade6c12ad4a793f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-09 10:23:51.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 1h 20m 21s","accounts_in_message":[],"_revision_number":20},{"id":"d1dbc9638338a4cf1249b3e996df76c5f9134861","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-09 10:48:30.000000000","message":"Patch Set 20:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/163/ : ABORTED","accounts_in_message":[],"_revision_number":20},{"id":"ef75913b4ec8bf807f1d4749d5447b54c663af16","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-01-09 18:21:26.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"52a965bc1c4f8d1396d271535ee1555278a13059","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-09 19:39:27.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4ae34b47312945c28eff4476f758cfc6 : FAILURE in 1h 16m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/124d7b8320f644f8982714e0ddce4c3f : RETRY_LIMIT in 16m 39s","accounts_in_message":[],"_revision_number":20},{"id":"7ceb74bd138161cf5e18bd82a969db0708e89089","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 21:21:22.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/178c6824735b4797be8f56ca7424c880\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4385795e94394e8d93a0a06167195bc0 : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b47a716376a48e7b4b4c0013abaee16 : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1afc8dd89b0847c4b6c9452adb20d867 : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ca72da333cdc7d40f42fe26ad4a2ffb8ad04807d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 23:38:58.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c26d55e038414cbeae7e49a96afd93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2f3a3570c942ba8e68515142e7a6a1 : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ec95586acd04769b5d2829c0931d0a5 : SUCCESS in 7m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ef8ab3a08864ba1bdd6df1d144ddb72 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2156ab87632948e4a77566836725f166 : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/dd391c054b754710b25bfe151d14a359 : SUCCESS in 59m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99e1db4cc7de458b8cabe986803a2ffd : SUCCESS in 57m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d47cfab803624bcfa67dbae02a5ff72f : SUCCESS in 1h 28m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0167621f1a1a442c8610045ed495ca03 : SUCCESS in 1h 07m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/535c6b0d22cb4de8b1c0fa7140a89cea : SUCCESS in 15m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5e999d21af23498f8a59c2e550dedb6d : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dfd6b9d0092c486fac4c9cc2b858dbfc : SUCCESS in 6m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a516ca771bac4650a65e9f37161d4234 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc1fbeb0e8604cda8f7aee43cacd14b7 : SUCCESS in 5m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aa430ae260c84789b6b16d64dbc53223 : SUCCESS in 7m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ba79a2903f84000bbdab5b156210289 : FAILURE in 47m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/722bad25ffe04c82ae6f60cc19607569 : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a8c58352c2ad45618bd5e07eeb327741 : SUCCESS in 2h 00m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c100d93d36cb42a6b3f5f32e0c4e1a58 : FAILURE in 3h 03m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/724856e859bf4c29a0f406f3500b6ca5 : SUCCESS in 1h 42m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1e6f9393b9254fbba673f29bf85fbc6c : SUCCESS in 2h 04m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3dfb3fcc104f43cd8af35701c3237502 : SUCCESS in 30m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bf1898ca303a422da626c392daac69eb : SUCCESS in 1h 16m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/818b4dc33c904e0f961d79cf8cc07614 : SUCCESS in 1h 25m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/22e07dda69614a1f8fba117865f491cb : SUCCESS in 43m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9aae260a208a4d31b79da4619b8dc2a1 : SUCCESS in 1h 57m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1a31a683ad0f43719c5b9eae57679c50 : FAILURE in 1h 54m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfc77d671a804867af08854aeffc54be : SUCCESS in 1h 08m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c0a063b7aa24aeabdf88ef95570b355 : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":20},{"id":"87770dfad92551d28d6ddfdd4dc7d5ecffd08d9c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-09 23:45:54.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_01_2026_23_39_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":20},{"id":"cee69961c8ab716597f55ef0ca717fe098242076","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-10 00:45:22.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-20-fc : SUCCESS in 57m\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-20-iscsi : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":20},{"id":"9018213793dc4b31c0e175113d0ff6ab20748e4e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-17 10:42:28.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"68c8eeab21beb4c8fe716ff3a78d69b025106ced","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-17 10:42:41.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 912058,20\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"f8745e801c9bef9bc45e218523d3febe4a1de4a0","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-18 05:48:38.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"e1fa67d1c620fdb9366f75b8614732e52baa1e92","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-18 05:48:51.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 912058,20\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"1507e5d38761f1d1fbfa6b8589373bacef31f9e7","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-18 08:51:18.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b7c1c611bcfdd761ab0d4aea22148605ac7aa89a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-18 08:51:31.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 912058,20\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"99b6bd5e7df39a506659a9a859f096e7aeb699a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-20 04:54:21.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"56f3ba84e190ac3f035406eac8644ea869638919","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-20 04:55:05.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/270afb7a731844de86b5ca07a41b9bf0 : NODE_FAILURE Node request 200-0000659579 failed in 0s","accounts_in_message":[],"_revision_number":21},{"id":"6035d60ac1d25b4c2ca97437461d27bebcafbcce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-20 04:56:36.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"6e9ea699d455cc125f380b5400aa793c213c20c5","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-20 04:57:36.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/38bf432685cd44a091736260b8f46387 : NODE_FAILURE Node request 200-0000659586 failed in 0s","accounts_in_message":[],"_revision_number":22},{"id":"9e0fc2f59be336bbbc6d4984886700e866c781e9","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-20 04:58:17.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"5657b8ca36e3be797ce2a19c973fbfd18b4c38ee","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-20 05:00:03.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"2304cdb6863f0294e9ec5223fc02d3b4668744c3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-20 05:02:05.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c9267a36ba724477aeaa1c71a365959b : RETRY_LIMIT in 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9c1a9ad142694ad48cc11f2b91d6448b : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":22},{"id":"e6a942e05c538eb616bc98368eea27de9c1d3573","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-20 05:05:24.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"64ac51be5b84cf6cf05129009793825dcbc3a2d8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-20 05:31:11.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":21},{"id":"a9ab475e266686653b4c197514a7b2b1e21c310a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-20 05:35:55.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":22},{"id":"bf94aca74653af5626d7952a53a7e0c3426deb9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 05:40:12.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b8e7cbfdd64b7da67adc6216dddb99\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85d2108e2286452d800288209338d7b6 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f13e7b6e176146d486f974d7a1c3611c : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1e2867b3a0964073af577dd6d068f7ad : SUCCESS in 22m 36s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"ab5e5b9cfb5b6f462224c3040ddc74420eb1a4d8","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-20 05:49:03.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- hpe-fc-driver build/f227c5173e8742bdb6a3d13bc6b7e313 : FAILURE in 33m 08s","accounts_in_message":[],"_revision_number":22},{"id":"07e097675a65dcb840d962f817a4929a1a78ea5b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-20 06:03:01.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 38m 34s","accounts_in_message":[],"_revision_number":22},{"id":"331d648deeb0cb30531d7ecb67bb23f51a5e6e5e","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-20 06:04:18.000000000","message":"Patch Set 22:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/912058 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"bad081ed0966df963d3b2f119b6fe9890f199a1f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-20 06:05:35.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/0f32dec91d684337ba7d59196f32cfed : FAILURE in 39m 43s","accounts_in_message":[],"_revision_number":22},{"id":"ab30e109260fd150cb8b768c1a919852ac17e5f8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-20 06:34:08.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/752d4f2a8a324362be1a73ae83e60b6b\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/7be12761072f42058666b957d986c94e : FAILURE in 1h 35m 26s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":22},{"id":"be61f32d9c54ae6d33e1bcb5870c814dfa8b7abd","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-20 07:28:26.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/5cb5a71ccecf4714adcd8afaed69c5c9\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/1ac4102786ff4df5a59e3e9094bf6847 : SUCCESS in 1h 36m 42s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/6767d1d90d814b3599d810576797e02d : SUCCESS in 2h 22m 11s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":22},{"id":"9f855086bf220114e3f695944de2019fbc70018e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-20 08:00:51.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"7295691f5d99e27e341de754ff5b069a176b1b3d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-20 08:24:51.000000000","message":"Patch Set 21:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3372/.","accounts_in_message":[],"_revision_number":21},{"id":"595450ef7b8d46902aeb6dc0d176a0e93e68ed60","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-20 08:30:07.000000000","message":"Patch Set 21:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-21/FC/ : UNSTABLE in 2m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-21/iSCSI/ : FAILURE in 3h 35m 26s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":21},{"id":"06f642294546d6e85956387323643fd1640e87ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 08:33:57.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78405c48cdef4b40a5e35a809b09796a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/747d1754e9f0478c937af3106f1919c8 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f2b68d3b9a448ef970622fb75238ec2 : SUCCESS in 9m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ae3fefa0ab94d4fa4cfc509dc332775 : SUCCESS in 10m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a697b15458524c3c95d885dc822d19c1 : SUCCESS in 14m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0261424815d4fef92a24da7fd75b6c8 : SUCCESS in 16m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/394a9da8ceb348da987c40479b51c0e7 : SUCCESS in 59m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d7113ec000e43d187a2cd19f91d02a0 : SUCCESS in 58m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a70ea7ab82a04d2f8dcbba87a143583e : SUCCESS in 1h 46m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41079ec972af4564af41a7807b17b729 : SUCCESS in 1h 00m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53f5cb52528f42dc8f7bd5638ec8f42c : SUCCESS in 14m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ea9aac685944049ba9f9809ab5ab4ae : SUCCESS in 22m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7ccd3a08f8346d6b5611580f53fba5e : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d614064743e7415eb3bac76f8ada7d65 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3a60a67c331e47159c4d1efcf129adc9 : SUCCESS in 6m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/618843ca51994cbba6cbadca404cf0e9 : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2babd57c21444affa09f490108b217c2 : FAILURE in 44m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e793e3cc6a24b07a64315d54e22d33c : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c01acdfe7600437e8369a88b001fd03f : SUCCESS in 1h 39m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bbbe0a68a18a49a68c39088f227c08a5 : FAILURE in 2h 05m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1164617e35ae461ba6f36ebf716e3760 : SUCCESS in 2h 14m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/977509bda11e4dcc8af846940a0074b5 : SUCCESS in 2h 07m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8e1cac1e827646089532ef3b4864d505 : SUCCESS in 37m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8d9dd44dd8ee469c87e1b4bd5df625e2 : SUCCESS in 1h 14m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1ffb6be9319486f9644b335693b6d61 : SUCCESS in 1h 38m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/62c8440b77704ef19c5f8ceea0d99399 : SUCCESS in 39m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/105d44d3ee6b41218f293b1f1a7c4017 : TIMED_OUT in 3h 34m 25s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3e5055f0008f49e8a4089692877007e8 : SUCCESS in 1h 29m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc15dd11e433411494f93a39695bac55 : SUCCESS in 1h 32m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d5d7fa9cecc4eee906740b470641b10 : SUCCESS in 41m 42s","accounts_in_message":[],"_revision_number":22},{"id":"a4359c7aeb350140ca10fafbbece782ce6e02e4d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-20 08:47:14.000000000","message":"Patch Set 22:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3373/.","accounts_in_message":[],"_revision_number":22},{"id":"820d84c80d71918f30d5b21362a36c47cdee2c1a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-20 08:47:18.000000000","message":"Patch Set 21:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3545/.","accounts_in_message":[],"_revision_number":21},{"id":"c4f7bff77761349bc2f839af50e6ce3320372400","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-20 09:30:08.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/94a180da2ed1476dbca2c8c70734db3c\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/b1e9f5632f9246a9b8843ddd6a1c1de2 : FAILURE in 1h 28m 55s\n- powerflex-v5-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/ff0867760c624ef1ac828f33d05cc889 : FAILURE in 1h 27m 44s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":22},{"id":"386d52cf23e9bb883b2c01c27e51a4824a7b71e7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-20 09:49:05.000000000","message":"Patch Set 22:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-22/FC/ : FAILURE in 4h 43m 12s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-912058-22/iSCSI/ : FAILURE in 4h 51m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":22},{"id":"2116da326c57eb2d8c39079b89101bb15b211a10","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-20 10:09:17.000000000","message":"Patch Set 22:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3546/.","accounts_in_message":[],"_revision_number":22},{"id":"62c85c8139a9c93de2fba2773961e928efc64ec9","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-20 11:16:26.000000000","message":"Patch Set 21:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1191/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":21},{"id":"8d44c8216217378d0cf710bf544a59ed2be2eaf3","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-20 12:54:19.000000000","message":"Patch Set 22:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1192/index.html : SUCCESS in 97m","accounts_in_message":[],"_revision_number":22},{"id":"15f2f53d0f5ab5e8f53a809bc37c3544255a99b8","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-03-20 18:15:24.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-21-fc : SUCCESS in 49m 55s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-21-iscsi : SUCCESS in 1h 1m 22s","accounts_in_message":[],"_revision_number":21},{"id":"8f3b7fefe79c1f6dc2ef69313142bb45357c03b0","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-03-20 21:09:43.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-22-fc : FAILURE in 1h 4m 1s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-22-iscsi : SUCCESS in 1h 1m 49s","accounts_in_message":[],"_revision_number":22},{"id":"229129a8eae416b9d8539f6fc3ab2bee5f62b7e1","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-03-20 22:31:39.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"cd9a5107ccf29f24a71fb3ae1e2832deb1cbc2d6","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-23 10:12:08.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"41370c49fc08da1a2d90a96569854db5d0d34515","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-03-23 12:55:32.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/7e1d1ce5c2f3424cbb07f7039eb766a9\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/06dc083b338c470e9cb4e872a724d755 : SUCCESS in 1h 42m 11s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/db6a7e312f57433db06571a43954abce : SUCCESS in 2h 42m 32s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":22},{"id":"ec0be2ce43a8b0bd52c8b4ab453d640713041e03","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-03-23 13:31:16.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"8218fdb6006a911a08f2090703d0515c8ae6872a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-04-20 11:24:05.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"d082ed45b4cdcd12028b084236d782b9f553690f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-04-20 11:27:01.000000000","message":"Patch Set 23:\n\n(4 comments)","accounts_in_message":[],"_revision_number":23},{"id":"5eb07783c76eeff960a8a7421a1ff43880a3ae3e","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-04-20 11:28:06.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"808b301eafa1dfa74ad49e4ce87e80352eaf214c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-20 11:28:46.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4796d23e9e9e43e7bf5c79bbf8b55665 : RETRY_LIMIT in 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ceb8d32adc944634baa5cfabb2e1b990 : RETRY_LIMIT in 26s","accounts_in_message":[],"_revision_number":23},{"id":"ca908dace25f1a3d042684db85187d0f695f38d7","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-04-20 11:29:08.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"b3ab5a910d24d3734ee3b489e40e245f00a02afd","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-20 11:47:39.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/9a50a0d726ec41c0a36ddbc8a6ddd2a9 : FAILURE in 23m 00s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/912058/etc/tempest_log.log","accounts_in_message":[],"_revision_number":23},{"id":"825c7005b280d34c6e61c70739199bd93c5e4940","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-20 12:01:16.000000000","message":"Patch Set 23:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":23},{"id":"65d00415797d5745edf0fd6f05c9ffee7cfe3eb8","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-20 12:02:44.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- hpe-fc-driver build/12399797f69e470ab3ec36429ab9d0aa : SUCCESS in 38m 01s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058/etc/tempest_log.log","accounts_in_message":[],"_revision_number":23},{"id":"b6a43662cc6e1c53c4488373d8514f752d07018d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-20 12:04:36.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912058 : FAILURE in 39m 22s","accounts_in_message":[],"_revision_number":23},{"id":"bdeacade4ab108a6cd764415b92d140c4814e88e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-20 12:07:50.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/522d99d891e24ed49cb6d744ed4ba27f : SUCCESS in 43m 13s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/912058/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/912058/etc/tempest_log.log","accounts_in_message":[],"_revision_number":23},{"id":"cbc6e4cf900648f66633c86dea31e7fa9e287968","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-20 12:49:41.000000000","message":"Patch Set 23:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3897/.","accounts_in_message":[],"_revision_number":23},{"id":"db1beb09e8d48bb5abaca19cf436c1c0693c67a0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-20 12:53:34.000000000","message":"Patch Set 23:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3721/.","accounts_in_message":[],"_revision_number":23},{"id":"834a95331e0be34b29f59f33d756c5a5c2899e59","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-04-20 13:32:53.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/ec1c7fb35c424769a6e4e71959650944\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/39ee7cf462464621813a984224678423 : SUCCESS in 1h 29m 21s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/bf94e22f13584d339d4b9b672c6ba7e8 : SUCCESS in 2h 02m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"98fe98cc2b6872b2044c17f50cc85a22cb8a70bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 13:40:47.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d8dba52db07496baa9a9eb8114c26a7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28f2c36eb49343dca0419ac0d37fac80 : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/43a64ce7c23747e498d3365dc0344f4f : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/df3bd31b0b894e32988efaeeb4a483bd : SUCCESS in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"9f86749534acdee400cfb2f938e459a69c105df8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-04-20 16:18:16.000000000","message":"Patch Set 23:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1554/ : ABORTED","accounts_in_message":[],"_revision_number":23},{"id":"a25ae3482e3a6ad35fe15d5899137ee161f70f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:11:03.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa4dde239d0840a68f26bf4844c2b2d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e49c7bd1364e709adab6f1b4f80379 : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72607eec9672477ca02845037f24294d : SUCCESS in 11m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bc51cb33c354328abf581b894fc8c9d : SUCCESS in 12m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a9d9215cf4b045e7887c36f68ebd187e : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab3b1da9f824b848c6b4e3c44546790 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/8eb8fd2f8463411ba92b742b85055480 : SUCCESS in 1h 07m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d1147557bf04617a05d1f4e7a744d34 : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/352a065a962842979ec72a26ff0d3da0 : SUCCESS in 1h 48m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c754490967c54ed38888f82cf2d528ea : SUCCESS in 1h 09m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef6bc8a0189c4f6ea1d6a2617749d601 : SUCCESS in 23m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1ba019bbeb774d0596e0c94cddb3808c : FAILURE in 26m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab77de2da635434a8fbc68d8bf129f28 : SUCCESS in 5m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f217e99af3245ee98c55f6d1a5ec8d3 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff928513b5ed4a20bd7ed7d99410fa9a : SUCCESS in 6m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ee481fb96bb244f1b12f2bcec37f42e0 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/00f7a51f4a024bf1be075acf553c6355 : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f01b27e6458405f82ecca793bf5050e : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5cc9b89035944669b6a54b38f9e407d : SUCCESS in 1h 51m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/58b083bdb5da47b58f859bd89d696083 : FAILURE in 2h 56m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7ec039617ef946968471d5929ae9f518 : SUCCESS in 2h 05m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d4804c967ca347f28daf130c54a511c1 : SUCCESS in 2h 00m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/db9c029ca8ab405c8cfaa2ed05eced84 : SUCCESS in 41m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/29ce60048f7a4adbb7848afef82695c1 : SUCCESS in 1h 16m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8fb4b756f9bf49439e6f3a210ff23731 : SUCCESS in 1h 44m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c34b6202304646b6969fa92b521c050e : SUCCESS in 27m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/37a4bded0b354529bb32fe3b0b800a23 : FAILURE in 2h 15m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71fc7830966a4171a448205aeccbdb3f : SUCCESS in 1h 34m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ac85ff0dee24c31a1896bca777f4377 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":23},{"id":"1723e1bded0aff3e95a56e1a374ae13ddb7a0192","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-04-29 06:06:55.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"29b2585025bf17c95805a41b63c16c91aae5c1a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 06:27:09.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/840faf1f86954bdf95e1e777a853e8b9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d21780601544b9cb161a2d984aecb96 : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aab89284d5754fc084936b53dbac8b9a : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a3d60a30684442f9a8a5bf95e071e0dc : FAILURE in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"d48fcfd06a628c047af8a185f24b380843f82b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 14:28:05.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f3e0621025468c8d9290fae07ee353\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaff171b310c4034bcff17289dc08f0d : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8a20632ebe4430ba1c74cea5360181d : SUCCESS in 11m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/839234eef1a749299c08858639c3a39e : SUCCESS in 10m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5367a35fa01345d996fa89570b235dd4 : FAILURE in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec44e1669fe44e2c9c611dc5595908a9 : SUCCESS in 17m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/acd318f8b1b54b2face95176863f8dc9 : SUCCESS in 33m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd6ba9c9c6394eb692926345dfbd64c9 : SUCCESS in 34m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc8969f72f074e8b8938933526e8110b : SUCCESS in 1h 40m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/267f0fa1d00942c09f4a4a29dadf7cee : SUCCESS in 33m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20367371b23f4b5ab4252830a981dcf7 : SUCCESS in 11m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/633d89c7439e441db7aea01d21ed18bf : FAILURE in 25m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fe898111ece2452c8f1ed2381efe931c : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a84297d4c1d4c8980275b373c9e6853 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a87179ff52b4408aab4c0454634e207 : SUCCESS in 6m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9c6a4e0f6e50497a8d96e3ad2c13a31a : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ac26bc0f929a4751bb47761efc256cec : SUCCESS in 28m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a141b67e40b40c19948c11bbcf3db64 : FAILURE in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/872ef701e62b4293a3c049e4b82bcaae : SUCCESS in 1h 45m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2669aca6e0d64c1bbd8344fb7f45ebae : FAILURE in 1h 32m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5598d2f97ef74125b9922e81b1d37384 : SUCCESS in 1h 48m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ca4921c2f7184e57adcfec34452366bb : SUCCESS in 1h 53m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b8477959ed354a2db64bc1ccafc44e3c : SUCCESS in 26m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/314f8d6717d143baa0bac2ecabe9cb09 : SUCCESS in 37m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/54340f98530348c7a44b8bfc9c9b1c1a : SUCCESS in 1h 35m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7756d609773e4a2caec54898da87e783 : SUCCESS in 27m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/33708bdf7e334be1b17a13e69f54b9d7 : SUCCESS in 2h 06m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c9111b037a724024a0d8fd1838794e0e : SUCCESS in 1h 39m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4717c383038f496ba180befba8da0e9c : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":23},{"id":"124395afd006494a2e07f43ce2b945ba3ffa0ca3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-29 14:37:25.000000000","message":"Patch Set 23:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_04_2026_14_31_19\nLeave a comment with","accounts_in_message":[],"_revision_number":23},{"id":"0112369a92c8eea1ac4ab5a4e5a2e4858e1b8f07","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-11 21:02:36.000000000","message":"Patch Set 23: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"5a38bbfd7ce837f6e7e668355090a740bcefcd17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-12 06:40:55.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"91a5a3f66d9981c3b0e18319787f36093c215632","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-12 06:42:28.000000000","message":"Patch Set 24:\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"611d76afe284f5fab28eedc80d3ed40bba889466","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-12 06:42:50.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"8e56888d2ecfec5e8abded064c168af75228493b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 07:05:59.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e79f35b5494d9ba973657f41ac508e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/607de2deb1574d27a83fe90cc263bf2f : FAILURE in 6m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc40243ab16645cd90901fcb66554e18 : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6658333f7925411cba11e410653abc28 : SUCCESS in 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"2a9bb335b5640089d06c0f21f6e3b06453236e56","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-12 07:12:59.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- hpe-fc-driver build/1a94118c169d4720b7c02f3b828c063c : FAILURE in 31m 33s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058/etc/tempest_log.log","accounts_in_message":[],"_revision_number":24},{"id":"20fc9a5171184f64d8933ad44a375f4f9373afd1","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-12 07:22:47.000000000","message":"Patch Set 24:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/912058 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"0267f8bc8cc8d1aa2ac2be66f6b25051a3df9ef6","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-12 07:33:33.000000000","message":"Patch Set 24:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/912058 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"0aa103844ee5a4ddfd08507361b66c95a48a0597","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-12 07:48:39.000000000","message":"Patch Set 24:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":24},{"id":"1d1d4a2fc63e8749da56e1108f1b07a4eebb761b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 08:49:58.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f9782b223f43b1837102e9f67bb842\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e0171e571784b3ea4c30106a0ef1cc0 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ee6a5ed4c6e4dfb9dd335077f54892e : SUCCESS in 9m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c9a8649d94a4edfa5b1d7026433636a : SUCCESS in 10m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7eb570415474792a6797511cc657097 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b6df44783234eb7bb502cd935d8f32d : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/b8fede8dda0249a1b8d05c76574d2b28 : SUCCESS in 35m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56734a3862cc4bc39b50f117bc4bc7ff : SUCCESS in 48m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a03c717ff504201b6846f51131085fe : SUCCESS in 1h 48m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19aeaee18fcd4b2399a96a06bcbb7094 : SUCCESS in 1h 01m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e51b5e9939484c76a7a21a02d89734d1 : SUCCESS in 14m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9bf5fa637a34946abd99e7da800c603 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2f166dcdc9e847cc967dce41cc5ae3d9 : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e35a6163d36c4e57b6794ed582e277d3 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9c61cd4e20f441a2be01087eb1198dbc : SUCCESS in 5m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/45114180f8894a389fd7ad87c504eda1 : SUCCESS in 5m 31s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/08f0f455c4964f80af71ec7fd86d22c3 : SUCCESS in 8m 51s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bd562b2f32eb40ac85638cbb24dcfb04 : SUCCESS in 42m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/87e7751fe3fb47c993b9ad9bf49bb6cb : FAILURE in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f83729dd9c4413c96587593a0a283b7 : SUCCESS in 1h 47m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/599761b060884926a353fbce2d963062 : FAILURE in 1h 53m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a5ab1ed171204f4dadac93b0e7dce881 : SUCCESS in 2h 03m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/792a3d0d1dc94e5fbef230194a0352a9 : FAILURE in 2h 02m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fbc9dcfe01f443ba83c28f93e5c840f2 : SUCCESS in 41m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/03dceee8bea943d698c285af5fd7733b : SUCCESS in 1h 18m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4003540511704337be121b776716180b : SUCCESS in 1h 46m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c6ea3e4dd9554ab48854a571d774b03e : SUCCESS in 26m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0288ca67392a4bf286e48febeaf0c1f9 : SUCCESS in 1h 53m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7368f92f3a5d4577aa8f900d6acadfa1 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":24},{"id":"565f8cffc9a0406919907f097307c92a07d93093","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-05-12 09:51:28.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/ebe62852c210417e9b5d409bc31c2e6d\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/d9fa31c4f27b47e3a43006e09b5e7c64 : SUCCESS in 1h 53m 53s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/bfc2a3f1fce84358a9bb011d71d50cec : FAILURE in 3h 07m 43s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":24},{"id":"bb667a288ffa886e9b021ae4e0aee6ebdac53381","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-12 10:27:49.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-24-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-24-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"081967364e2b6a3dc713d821e3b2b63a298c21e0","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-12 10:44:29.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"34fff2798aef3b57ac8535b0a780bed21e5918ab","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-12 12:47:17.000000000","message":"Patch Set 24: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"893c29c3667937762c4f339730b583a3906244e4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-05-12 13:08:42.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/192dc42d60ce453eb194e42ad0aa7fca\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/684633e762c44fe098c3940f707b0e95 : SUCCESS in 1h 55m 04s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/49b3a88d38634797b1a55e89af34600b : SUCCESS in 2h 23m 15s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":24},{"id":"ef76ed73ddf2a04b7c4629a50de4d7e825a69773","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-12 13:18:49.000000000","message":"Patch Set 24: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"7c654d940bafce83eeb347e3d7fc48f7210fe997","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-12 14:56:57.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"a978da909a70266124efd435c51af59387d47f1b","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-12 15:00:12.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"b3838ccf8fc08b4a6d756918d879401bf0ce245a","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-12 15:00:46.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"3ce79e8181ba791c4b85671c78b1e785d1f3ed6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:13:02.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331ef743a6df41719e13387bd513ae60\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3cd69b3f0c7847b79365bae3037bd2e5 : FAILURE in 4m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/330bd92546a84cc8bc9c07881631dc24 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/819a404380f14ce4b060c48ba671cff5 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"b56c0c8935e91f40f3db7d20a9e528d5517536aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-12 15:57:09.000000000","message":"Patch Set 25:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic22add411b594d90d12b94e85055a33db40c197d .","accounts_in_message":[],"_revision_number":25},{"id":"f295afc556513cb04a729b6cd9a1669fa7b622cf","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-12 17:34:40.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- hpe-fc-driver build/478bc0e771d44154b8a6b046bbf8179d : FAILURE in 43m 58s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912058/etc/tempest_log.log","accounts_in_message":[],"_revision_number":25},{"id":"4d29c19ea7c267ce58e4ec27d8406f2990297ff7","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-05-12 17:34:47.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/307c858566684947b00ca75b6a793312\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/2c71db7ad9ea4f018e0f813b9ffb2c6c : SUCCESS in 1h 35m 11s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/9b0889399df8439aad4ef5afde1669bb : SUCCESS in 2h 33m 35s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":25},{"id":"cad3e2a9cb0011e193455114651175e87e4887a0","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-12 17:59:00.000000000","message":"Patch Set 25:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/912058 : SUCCESS","accounts_in_message":[],"_revision_number":25},{"id":"5700d3f049202e19dfae6f275ec2e7e2d15bbbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 21:42:33.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ea3659bc764ae085e1bd275e26f327\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b823aec8daaf4d93ad43e61b5c7cd15e : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d34a7f00014a4a8a934684c13c2abde5 : SUCCESS in 12m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aeefe4db4e8f418c9ed04865a4d52d76 : SUCCESS in 11m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/819c326c39f34cf8a24713ffc2915a58 : SUCCESS in 17m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5a66f6f81142709c4fc3841155935a : SUCCESS in 14m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/e46aa989c1f84b839850b78684448c85 : SUCCESS in 58m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efbaca408d484cf591862c7937093b32 : SUCCESS in 1h 07m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b39e63fa2022462281fa599e5aaca478 : SUCCESS in 1h 41m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58b55f081b14451ba6c07c89e95f58a6 : SUCCESS in 1h 13m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72a158fded3a48d8a229a0ab91abac6c : SUCCESS in 21m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/03c2acdfbb264c7ba7ccdbc977573fb4 : TIMED_OUT in 41m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2e921fe2a4324211bd263eb58514bb0c : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/911988367a044cf68aef9094e2cbfeff : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/84c13c5ac2ec440382eae24d5cace28d : SUCCESS in 8m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/269629ac67bc4d4c8320326be4b1e4e5 : SUCCESS in 6m 12s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e47cbbbf3c714a5dafd131dd93d6c92d : SUCCESS in 11m 13s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/68531f8555454021aeb953340e1bb16d : SUCCESS in 47m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ab5557face24272b4821e188b534ccf : FAILURE in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3256bcc975924fec866e5d174048cf8b : SUCCESS in 2h 04m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f522cb00021440009364ed2b6aa80ded : FAILURE in 2h 41m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e4d39c96a9084019b7d113fc3f692577 : SUCCESS in 1h 53m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ed0ccba419334a82a963b9b6070be108 : SUCCESS in 1h 34m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/124ae93f2f764109a1333638986ac1c7 : SUCCESS in 27m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dc1fd0e76b4d4f4aa1d2c75bb71ea6aa : SUCCESS in 1h 17m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/28c3fe1db82c43039d11f9a65a1bf379 : SUCCESS in 1h 47m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a7269685e0045edb95d6ab06cd6d49e : SUCCESS in 31m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/870f6d31be884c09a74318c7fbaf2af8 : SUCCESS in 1h 53m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4305d39c177423c9af3efa3194b54e7 : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":25},{"id":"e6b77e28cc61216b00089e0ae9e3cda4327caf6b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-12 21:52:34.000000000","message":"Patch Set 25:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_05_2026_21_45_51\nLeave a comment with","accounts_in_message":[],"_revision_number":25},{"id":"9a20d708007ee07ed99e3a0e860ba23d8709f024","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-12 21:53:26.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-25-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-912058-25-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"b5f3012f2226be33376ab599e4183ca1f78c3e72","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-13 13:45:01.000000000","message":"Patch Set 25: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"d41d64dc9485dbb308e12fbc2b9863272f4991f6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-13 15:15:21.000000000","message":"Patch Set 25: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":25},{"id":"e2f3edc1679fcc473d05bd262712739ddd1bafc2","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-13 15:37:18.000000000","message":"Patch Set 25:\n\n(5 comments)","accounts_in_message":[],"_revision_number":25},{"id":"7cb91700332610bfb75c271f9acfc475d20e5ec5","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-13 15:42:05.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"d7fe49a204873a9986d0ed320ce77bd13723b3c7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-13 19:02:57.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"2266607fafde8c23245d3150d9c8110bc59c11e7","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-13 19:39:44.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"3f04730779266f1442268f3262dd09328a603b33","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-13 19:52:38.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"f8adc92c45f65bbea5199768824ac1bff46c480a","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-13 19:57:54.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"b0ff925ec7754e8f2e425736bee81841b0eb0c05","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-13 20:53:43.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":25},{"id":"0c8ec23bf6892143a015b0ce5e06646b8d343073","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 20:54:37.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"7eee4c8f4f58a8458d94e2c3e5ceae6e3b9cc857","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-13 21:02:14.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"27a17fb8696c726a6b3e269bf3fd73ba76be4da2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 22:52:59.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2926be212c149c8baa2b8acccfb30a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/396d6c760ee14f04b94260d16dc5eca4 : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d09042bddbf74d35a5efa939f51e7993 : SUCCESS in 12m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4647a45d8b34427ad3afbad11a8b4e4 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d94a8941ac45db99466c568983d367 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/40ed3e56595e4006a2a98830221a8906 : SUCCESS in 32m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48a621746fbc4ba880272280bdffbece : SUCCESS in 56m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea0c07fe932947ce8964ad5c2476ab2b : SUCCESS in 1h 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b88dc78c6ae84343b8c9de44fc184f5c : SUCCESS in 32m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60f4fc973c6f4e7e864638f9d0cd1b69 : SUCCESS in 6m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4e7593e81b634b2897adfa8424f02bdb : SUCCESS in 1h 04m 43s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c63593829f3c4e50a1755acd36d0f17e : SUCCESS in 1h 53m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e3845f3b027496ca0e6b24d0f93fd15 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":25},{"id":"fd354a90f40a8bc6b5890c02098c2a33d2303ff9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 22:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":25},{"id":"0b31d7467ad5880f289b9a4d969f69276ca5afc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 22:54:14.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79d908894dce4e248ff37cf9d691fd41\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b4091729574cd09543298fa2e6ca55 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c3e79d3ad054ba983fc2b5777aebb0b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":25},{"id":"17c26700b94c6006fe8251d46d2e281c3e3edece","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-16 21:19:46.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/8bf39d51d46c44ac879d6d6de67f21df : FAILURE in 28m 51s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/912058/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/912058/etc/tempest_log.log","accounts_in_message":[],"_revision_number":25},{"id":"f64ec7feef997d7b89437c0f95a622b4aade0ba0","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-21 06:02:59.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"9c9f9202df6cc69abc47a25f7a96867c9d74c6ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:51.000000000","message":"Patch Set 25:\n\nBuild Failed","accounts_in_message":[],"_revision_number":25},{"id":"e4080cfec343ed90dedd3f3a0ee4736f22844e0c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:05:40.000000000","message":"Patch Set 24:\n\nBuild Failed","accounts_in_message":[],"_revision_number":24},{"id":"bb6781633b426af320315b39e225fba149031e38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:06:18.000000000","message":"Patch Set 25:\n\nBuild Failed","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"24a0db57e9c665dc3c23ee7c6195c42e3fae06d6","revisions":{"f2c57467c4654be7464a314dbb22c199fe4b2bf5":{"kind":"REWORK","_number":1,"created":"2024-03-14 04:54:31.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/1"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-14 04:51:56.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-oF support","message":"Dell PowerFlex: add NVMe-oF support\n\nThis patch adds NVMe-oF support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2c57467c4654be7464a314dbb22c199fe4b2bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2c57467c4654be7464a314dbb22c199fe4b2bf5"}]},"branch":"refs/heads/master"},"213352009026dc807763755e33282520e33dc3a7":{"kind":"REWORK","_number":2,"created":"2024-03-14 07:04:55.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/2"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-14 06:50:07.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-oF support","message":"Dell PowerFlex: add NVMe-oF support\n\nThis patch adds NVMe-oF support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/213352009026dc807763755e33282520e33dc3a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/213352009026dc807763755e33282520e33dc3a7"}]},"branch":"refs/heads/master"},"198405c7c2b862351d1470e2a3ba3554e7e811cf":{"kind":"REWORK","_number":3,"created":"2024-03-22 08:10:35.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/3"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-22 08:10:24.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-oF support","message":"Dell PowerFlex: add NVMe-oF support\n\nThis patch adds NVMe-oF support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/198405c7c2b862351d1470e2a3ba3554e7e811cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/198405c7c2b862351d1470e2a3ba3554e7e811cf"}]},"branch":"refs/heads/master"},"40b33523c555bea1451b1432e9515a3cbc45856b":{"kind":"REWORK","_number":4,"created":"2024-04-02 04:58:29.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/4"}}},"commit":{"parents":[{"commit":"54856da91045299537fdb69edf43fb61aba79cc6","subject":"Merge \"PowerMax: Allow live migration without pool name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54856da91045299537fdb69edf43fb61aba79cc6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-04-02 04:58:24.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-oF support","message":"Dell PowerFlex: add NVMe-oF support\n\nThis patch adds NVMe-oF support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40b33523c555bea1451b1432e9515a3cbc45856b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40b33523c555bea1451b1432e9515a3cbc45856b"}]},"branch":"refs/heads/master"},"04ddcc5ba5e28edea549e8bb876d6a1574b2d455":{"kind":"REWORK","_number":5,"created":"2024-04-07 06:19:19.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/5"}}},"commit":{"parents":[{"commit":"54856da91045299537fdb69edf43fb61aba79cc6","subject":"Merge \"PowerMax: Allow live migration without pool name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54856da91045299537fdb69edf43fb61aba79cc6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-04-07 06:18:39.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-oF support","message":"Dell PowerFlex: add NVMe-oF support\n\nThis patch adds NVMe-oF support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04ddcc5ba5e28edea549e8bb876d6a1574b2d455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04ddcc5ba5e28edea549e8bb876d6a1574b2d455"}]},"branch":"refs/heads/master"},"c5264d0b187945cf6c308aa4ff08f3cb28815551":{"kind":"REWORK","_number":6,"created":"2024-04-07 13:56:06.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/6"}}},"commit":{"parents":[{"commit":"54856da91045299537fdb69edf43fb61aba79cc6","subject":"Merge \"PowerMax: Allow live migration without pool name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54856da91045299537fdb69edf43fb61aba79cc6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-04-07 13:55:37.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-oF support","message":"Dell PowerFlex: add NVMe-oF support\n\nThis patch adds NVMe-oF support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5264d0b187945cf6c308aa4ff08f3cb28815551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5264d0b187945cf6c308aa4ff08f3cb28815551"}]},"branch":"refs/heads/master"},"0d189489bd5e6e282dd4c0b51eef901a326908e6":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-04-11 13:50:23.000000000","uploader":{"_account_id":36952,"name":"Karam Jameel Moore","display_name":"KaramJameelMoore","email":"wawmart@me.com","username":"KaramJameelMoore"},"ref":"refs/changes/58/912058/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/7"}}},"commit":{"parents":[{"commit":"54856da91045299537fdb69edf43fb61aba79cc6","subject":"Merge \"PowerMax: Allow live migration without pool name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54856da91045299537fdb69edf43fb61aba79cc6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Karam Jameel Moore","email":"wawmart@me.com","date":"2024-04-11 13:50:23.000000000","tz":0},"subject":"kingkaram.com Dell PowerFlex: add NVMe-oF support","message":"kingkaram.com\nDell PowerFlex: add NVMe-oF support\n\nThis patch adds NVMe-oF support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d189489bd5e6e282dd4c0b51eef901a326908e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d189489bd5e6e282dd4c0b51eef901a326908e6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c3fa549746ba1dba18d3000198cc26b9127838c9":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2024-04-11 13:59:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/58/912058/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/8"}}},"commit":{"parents":[{"commit":"54856da91045299537fdb69edf43fb61aba79cc6","subject":"Merge \"PowerMax: Allow live migration without pool name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54856da91045299537fdb69edf43fb61aba79cc6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-04-11 13:59:17.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-oF support","message":"Dell PowerFlex: add NVMe-oF support\n\nThis patch adds NVMe-oF support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3fa549746ba1dba18d3000198cc26b9127838c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3fa549746ba1dba18d3000198cc26b9127838c9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7f1753cc64b0c6e63e83ae76a9de177cc503c1f7":{"kind":"REWORK","_number":9,"created":"2024-08-19 09:52:11.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/9"}}},"commit":{"parents":[{"commit":"24a560f4c05e253a21e28cd496b299db8b503eca","subject":"Merge \"gcs: Remove unused fallback to oauth2client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a560f4c05e253a21e28cd496b299db8b503eca"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-08-19 09:51:55.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f1753cc64b0c6e63e83ae76a9de177cc503c1f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f1753cc64b0c6e63e83ae76a9de177cc503c1f7"}]},"branch":"refs/heads/master"},"4f3c6b6ec9932e0bfa627eae7b166a8b225b2ad7":{"kind":"REWORK","_number":10,"created":"2024-08-20 03:45:12.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/10"}}},"commit":{"parents":[{"commit":"24a560f4c05e253a21e28cd496b299db8b503eca","subject":"Merge \"gcs: Remove unused fallback to oauth2client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a560f4c05e253a21e28cd496b299db8b503eca"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-08-20 03:44:47.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f3c6b6ec9932e0bfa627eae7b166a8b225b2ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f3c6b6ec9932e0bfa627eae7b166a8b225b2ad7"}]},"branch":"refs/heads/master"},"c9689acd5b705365d736b33f05cffebee3254049":{"kind":"REWORK","_number":11,"created":"2024-08-22 12:47:02.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/11"}}},"commit":{"parents":[{"commit":"24a560f4c05e253a21e28cd496b299db8b503eca","subject":"Merge \"gcs: Remove unused fallback to oauth2client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a560f4c05e253a21e28cd496b299db8b503eca"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-08-22 12:45:53.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9689acd5b705365d736b33f05cffebee3254049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9689acd5b705365d736b33f05cffebee3254049"}]},"branch":"refs/heads/master"},"bca14743cbf7dcfd9283ba77120a025217cc43a6":{"kind":"REWORK","_number":12,"created":"2024-08-22 13:10:31.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/12"}}},"commit":{"parents":[{"commit":"24a560f4c05e253a21e28cd496b299db8b503eca","subject":"Merge \"gcs: Remove unused fallback to oauth2client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a560f4c05e253a21e28cd496b299db8b503eca"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-08-22 13:10:11.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bca14743cbf7dcfd9283ba77120a025217cc43a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bca14743cbf7dcfd9283ba77120a025217cc43a6"}]},"branch":"refs/heads/master"},"f81ec929abc73deeec106dbece1c11050dfb10f3":{"kind":"REWORK","_number":13,"created":"2024-08-31 03:59:03.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/13"}}},"commit":{"parents":[{"commit":"24a560f4c05e253a21e28cd496b299db8b503eca","subject":"Merge \"gcs: Remove unused fallback to oauth2client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a560f4c05e253a21e28cd496b299db8b503eca"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-08-31 03:57:27.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f81ec929abc73deeec106dbece1c11050dfb10f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f81ec929abc73deeec106dbece1c11050dfb10f3"}]},"branch":"refs/heads/master"},"292c2f103d86771f77339dec16cf9cfea6ccf50a":{"kind":"REWORK","_number":14,"created":"2024-08-31 09:49:41.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/14"}}},"commit":{"parents":[{"commit":"24a560f4c05e253a21e28cd496b299db8b503eca","subject":"Merge \"gcs: Remove unused fallback to oauth2client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a560f4c05e253a21e28cd496b299db8b503eca"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-08-31 09:48:00.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/292c2f103d86771f77339dec16cf9cfea6ccf50a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/292c2f103d86771f77339dec16cf9cfea6ccf50a"}]},"branch":"refs/heads/master"},"2fdc2b84ee0afbe31ae5e72f05d24513d2376c0d":{"kind":"REWORK","_number":15,"created":"2025-01-26 04:49:28.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/15"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-01-26 04:49:23.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2fdc2b84ee0afbe31ae5e72f05d24513d2376c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2fdc2b84ee0afbe31ae5e72f05d24513d2376c0d"}]},"branch":"refs/heads/master"},"f0afc3b46227e073360b8e239dd352f6f5f4aea3":{"kind":"REWORK","_number":16,"created":"2025-09-30 05:30:10.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/16"}}},"commit":{"parents":[{"commit":"8064195521f9f511f630ff63b5dde1e88ff53e18","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8064195521f9f511f630ff63b5dde1e88ff53e18"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-09-30 05:29:25.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0afc3b46227e073360b8e239dd352f6f5f4aea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0afc3b46227e073360b8e239dd352f6f5f4aea3"}]},"branch":"refs/heads/master"},"baa5a5a6e3d54e658bb5ef460ff543bac49323ae":{"kind":"REWORK","_number":17,"created":"2025-09-30 07:44:32.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/17"}}},"commit":{"parents":[{"commit":"8064195521f9f511f630ff63b5dde1e88ff53e18","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8064195521f9f511f630ff63b5dde1e88ff53e18"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-09-30 07:43:02.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/baa5a5a6e3d54e658bb5ef460ff543bac49323ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/baa5a5a6e3d54e658bb5ef460ff543bac49323ae"}]},"branch":"refs/heads/master"},"6421669833f88feda25fee2ce048d1f93a9a8208":{"kind":"REWORK","_number":18,"created":"2025-10-20 06:25:47.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/58/912058/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/18"}}},"commit":{"parents":[{"commit":"8064195521f9f511f630ff63b5dde1e88ff53e18","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8064195521f9f511f630ff63b5dde1e88ff53e18"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-10-20 06:24:58.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6421669833f88feda25fee2ce048d1f93a9a8208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6421669833f88feda25fee2ce048d1f93a9a8208"}]},"branch":"refs/heads/master"},"e65088b40c85d36b36b1c6f13d99e032a01e5e77":{"kind":"REWORK","_number":19,"created":"2026-01-08 08:32:16.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/912058/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/19"}}},"commit":{"parents":[{"commit":"8064195521f9f511f630ff63b5dde1e88ff53e18","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8064195521f9f511f630ff63b5dde1e88ff53e18"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-01-08 08:32:16.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e65088b40c85d36b36b1c6f13d99e032a01e5e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e65088b40c85d36b36b1c6f13d99e032a01e5e77"}]},"branch":"refs/heads/master"},"f2b09161cb86262ebfad79ee0766cdf70dd44f95":{"kind":"REWORK","_number":20,"created":"2026-01-09 06:37:17.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/912058/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/20"}}},"commit":{"parents":[{"commit":"8064195521f9f511f630ff63b5dde1e88ff53e18","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8064195521f9f511f630ff63b5dde1e88ff53e18"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-01-09 06:37:16.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2b09161cb86262ebfad79ee0766cdf70dd44f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2b09161cb86262ebfad79ee0766cdf70dd44f95"}]},"branch":"refs/heads/master"},"b3ff15b9da6ca26d7d8ba44431e7be2c1ee0627e":{"kind":"REWORK","_number":21,"created":"2026-03-20 04:54:21.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/912058/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/21"}}},"commit":{"parents":[{"commit":"ddbffaa0ba328a5ebf53935a5b9c3659b499e4b9","subject":"Merge \"[Hitachi] Update docs for VSP One Block High End.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ddbffaa0ba328a5ebf53935a5b9c3659b499e4b9"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-03-20 04:52:33.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3ff15b9da6ca26d7d8ba44431e7be2c1ee0627e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3ff15b9da6ca26d7d8ba44431e7be2c1ee0627e"}]},"branch":"refs/heads/master"},"72d2de8ce2ffeb8f51ebcd99747731f70475b5f5":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-03-20 04:56:36.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/912058/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/22"}}},"commit":{"parents":[{"commit":"fcdecce8b40ee46e7197dba25aa095900baa26a0","subject":"Merge \"Dell PowerFlex: Fix san_username rotation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fcdecce8b40ee46e7197dba25aa095900baa26a0"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-03-20 04:56:36.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72d2de8ce2ffeb8f51ebcd99747731f70475b5f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72d2de8ce2ffeb8f51ebcd99747731f70475b5f5"}]},"branch":"refs/heads/master","description":"Rebase"},"5412f203cab3db87efa63ab9273360a7293fce12":{"kind":"REWORK","_number":23,"created":"2026-04-20 11:24:05.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/912058/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/23"}}},"commit":{"parents":[{"commit":"fcdecce8b40ee46e7197dba25aa095900baa26a0","subject":"Merge \"Dell PowerFlex: Fix san_username rotation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fcdecce8b40ee46e7197dba25aa095900baa26a0"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-04-20 11:22:34.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5412f203cab3db87efa63ab9273360a7293fce12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5412f203cab3db87efa63ab9273360a7293fce12"}]},"branch":"refs/heads/master"},"7d5f250e1a43b8a7d25761bcb843e5a42cccba6b":{"kind":"REWORK","_number":24,"created":"2026-05-12 06:40:55.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/912058/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/24"}}},"commit":{"parents":[{"commit":"2e906477168069f9620f854eefde6b76f4692679","subject":"Merge \"Updates tests to cover all NVMe-oF protocol variants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e906477168069f9620f854eefde6b76f4692679"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-12 06:40:48.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d5f250e1a43b8a7d25761bcb843e5a42cccba6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d5f250e1a43b8a7d25761bcb843e5a42cccba6b"}]},"branch":"refs/heads/master"},"24a0db57e9c665dc3c23ee7c6195c42e3fae06d6":{"kind":"REWORK","_number":25,"created":"2026-05-12 14:56:57.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/912058/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/912058/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/912058/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/912058/25"}}},"commit":{"parents":[{"commit":"2e906477168069f9620f854eefde6b76f4692679","subject":"Merge \"Updates tests to cover all NVMe-oF protocol variants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e906477168069f9620f854eefde6b76f4692679"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-03-05 12:48:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-12 14:56:39.000000000","tz":0},"subject":"Dell PowerFlex: add NVMe-TCP support","message":"Dell PowerFlex: add NVMe-TCP support\n\nThis patch adds NVMe-TCP support in Dell PowerFlex driver.\n\nImplements: blueprint dell-powerflex-nvmeof\nChange-Id: Ic22add411b594d90d12b94e85055a33db40c197d\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a0db57e9c665dc3c23ee7c6195c42e3fae06d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a0db57e9c665dc3c23ee7c6195c42e3fae06d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
