)]}'
{"id":"openstack%2Fcinder~912060","triplet_id":"openstack%2Fcinder~master~I9e1699632e7c835df0f200d7eba03c5472193500","project":"openstack/cinder","branch":"master","topic":"retype","attention_set":{"37328":{"account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"last_update":"2025-08-20 14:07:48.000000000","reason":"Someone else replied on the change"},"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2025-12-19 10:12:44.000000000","reason":"\u003cGERRIT_ACCOUNT_36538\u003e replied on the change","reason_account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}},"36897":{"account":{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},"last_update":"2025-06-09 15:15:10.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-12-26 08:49:36.000000000","reason":"\u003cGERRIT_ACCOUNT_36538\u003e replied on the change","reason_account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}},"37923":{"account":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"last_update":"2026-09-30 11:44:10.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"38067":{"account":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"last_update":"2026-09-30 11:44:10.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"14806":{"account":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"last_update":"2025-05-14 15:12:15.000000000","reason":"\u003cGERRIT_ACCOUNT_36538\u003e replied on the change","reason_account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-09-30 14:32:09.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-30 11:44:10.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-08-20 13:53:54.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"24005":{"account":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"last_update":"2024-04-02 12:54:02.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I9e1699632e7c835df0f200d7eba03c5472193500","subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","status":"NEW","created":"2024-03-14 11:28:30.000000000","updated":"2026-09-30 14:58:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":56,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c94625f1788b5e9d3ef8d6abdc99aa56ef132c9f","_number":912060,"virtual_id_number":912060,"owner":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-30 14:20:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 11:28:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 11:34:42.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-14 11:46:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 12:25:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 12:27:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-14 12:52:41.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 13:13:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-14 13:39:27.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 13:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 14:18:07.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-14 16:42:24.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 02:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-15 04:01:36.000000000","updated_by":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"reviewer":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"state":"REVIEWER"},{"updated":"2024-03-15 10:09:07.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-15 12:29:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-18 07:45:00.000000000","updated_by":{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},"reviewer":{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},"state":"REVIEWER"},{"updated":"2024-03-20 10:32:48.000000000","updated_by":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"reviewer":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"state":"CC"},{"updated":"2024-03-21 20:22:13.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-27 14:12:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-02 12:54:02.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-09-18 14:03:32.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 15:46:09.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-18 15:46:52.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-05-14 17:46:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-22 14:08:53.000000000","updated_by":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"CC"},{"updated":"2025-07-25 06:30:08.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"CC"},{"updated":"2025-08-07 16:56:07.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-08-18 14:43:26.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-08-29 04:36:46.000000000","updated_by":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"reviewer":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"state":"REVIEWER"},{"updated":"2025-10-30 19:44:58.000000000","updated_by":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"reviewer":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"state":"REVIEWER"},{"updated":"2025-11-07 16:15:09.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"CC"},{"updated":"2026-03-02 09:42:23.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2026-03-02 09:42:23.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2026-03-06 19:11:46.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"9e6117c3cd221700a86a7d987241a0d8157ccdd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-14 11:28:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62dee9bfb776ec123f06d3c41c585ed8bfe865c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-14 11:28:37.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16637/","accounts_in_message":[],"_revision_number":1},{"id":"779cf5f1a533fe4e6af8d7cae142452c67fb9930","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-14 11:28:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16637/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9097ac1c161169109f310e6ede32f35fb623bac7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-14 11:34:42.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/de2308ad0a9c42ecac52e79aea444780\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/54616c519e614da5b2436fabfd595b0b : NODE_FAILURE Node request 200-0000002767 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"0a2d3b2fdb049227154381b18e8b4bade8a313fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 11:46:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de4d187736fd42bb87c8d494e26e0640\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6025b2e8c8c7480db733e18fae4b154a : FAILURE in 8m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38ff832a02cf48e9b718ed5d8b970b44 : FAILURE in 9m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5386c7f4c208449fa879f85f858f9172 : FAILURE in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f1f7c31c83f49f036f517b0c557a1633ab7ea4b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-14 12:25:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0f7825486cb2e11c94408285a06bb30a6a4a624f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-14 12:27:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 56m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1b2e850a376448ef146c57afb629cb464b53e89b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-14 12:52:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/912060/1/ : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"52bbc9562b7c32869160c433b08b72f6882ac1ba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-14 13:13:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/912060/1/check/DellEMC_Unity_FC/aa44d8a/DellEMC_Unity_FC/1500 : SUCCESS in 1h 30m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/912060/1/check/DellEMC_Unity_iSCSI/ab3997b/DellEMC_Unity_iSCSI/1532 : SUCCESS in 1h 44m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6cd0742b722f13cdb50ae2dcd4a7aaa0dd54db0d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-14 13:23:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 52m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a9ff8642274c61aa6bf9dbb15cd293e5c87d82d4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-14 13:39:27.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/912060/1/check/DellEMC_PowerFlex_3.6/e035a94/DellEMC_PowerFlex_3.6/1641 : SUCCESS in 1h 49m 21s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/60/912060/1/check/DellEMC_PowerFlex_4.5/e001507/DellEMC_PowerFlex_4.5/580 : FAILURE in 2h 10m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"cc9b3b7f26132ec29a502b5c2d622628ea67b26d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 13:59:44.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/deb1389fc5144d2abf503b8c89290d97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4567f13987b64a2994de5200ae8b9765 : FAILURE in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0866da220df1496c903507ac710b28c4 : FAILURE in 11m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50e77b04360a48efb0d9d2033cce4679 : FAILURE in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196aad9d6702447087f5159aeb655167 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/e1b1b660c5dc46c6b0349cbef771d95d : SUCCESS in 55m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4ffe7513babd47a088bfe9ab26efcf11 : SUCCESS in 56m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8be25aaddf0448d96d9fa5fdf414709 : SUCCESS in 1h 16m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e37efb1d9bf4f32be54cbadb64a81c3 : SUCCESS in 53m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc5c5c44871c4f19bf03df8f68d3c36f : FAILURE in 11m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95c5fa4cd88c4ef597cd34d5957ef617 : FAILURE in 12m 39s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/902a26eb298f4048b6fab3966c648ecf : FAILURE in 16m 27s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5eff8efb272445b2acab3571b2a7c38d : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ae49389b6ff74152b6c02bc9d81c02e2 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/547bbcce36214b899fd6f867e283f599 : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca80e57c0a324c838402a7589d8b9b97 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a6efcecb7eac4bc89978ca559d137b93 : SUCCESS in 41m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d3264cbbf44d493ab47ed4f5078887bd : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d59a3a7efd9c411e8cf9df29eb29e092 : SUCCESS in 2h 02m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ed5f56c0cf71492ebffb4b27e78caeeb : SUCCESS in 2h 09m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/75c2950e251942df9f09d53f206b15e4 : SUCCESS in 1h 33m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/969e22169e3b431b985f53ab310506ec : SUCCESS in 1h 45m 54s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b066d8a6f55d4a7b8ca3ff5ceb5a1a7c : SUCCESS in 1h 09m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/11c207da097a4173af63c33be4d6c05c : SUCCESS in 1h 15m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/de535ed2323844d3aba7814cb3fbb8c3 : SUCCESS in 52m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8797c706fcbd4877810577947081f34d : SUCCESS in 1h 12m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/35eb690bfbad46ba87d9a551a8f47c96 : SUCCESS in 1h 22m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/69e88b004b3c4fcd9e6c03a99476ae73 : SUCCESS in 1h 48m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90005b616e624fd690efa542b947f49a : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e263d06c2ef6e945fbff0ea87d5fec1fb94a3063","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-14 14:18:07.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-1/FC/ : SUCCESS in 2h 26m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-1/iSCSI/ : SUCCESS in 2h 5m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"078c788f2a63e3d49d9693ed3baaa950e5149549","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-14 16:42:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/60/912060/1/check/fujitsu-eternus-dx-iscsi/cb4b42f : FAILURE in 2h 41m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/60/912060/1/check/fujitsu-eternus-dx-fc/0fca003 : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c0e69784c099a976a244086be31db5199e87f82f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-15 02:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/1/fc : SUCCESS in 0h 53m 42s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/1/iscsi : SUCCESS in 0h 50m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2ce0682b6985abc99670abbe5341c447d79b4473","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2024-03-15 04:01:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3c106fca74ea718f1310d5daadb87330cfe3205f","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2024-03-15 06:05:02.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"011283799ccc0ba3cf307c88bcbc4704915618d8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-15 10:09:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/1/20240315154212/NEC-V-ISCSI : SUCCESS in 3h 23m 18s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/1/20240315154212/NEC-V-FC : SUCCESS in 3h 26m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8e415536a375f59ba5b141461cad37c832479cb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-15 10:24:33.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2364c8b590358e7d201ccd7497c3c13d2f2454ff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 10:24:45.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16642/","accounts_in_message":[],"_revision_number":2},{"id":"4d88af77b8e9ebedd7182ecc78e3ebec448afa6e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 10:24:56.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16642/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d4ac53d933bdfb1c14d4c24471f17709af0db510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-15 10:31:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ac819792b22730c0e9af4a265d09dcfcbc2acb4a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 10:31:19.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16643/","accounts_in_message":[],"_revision_number":3},{"id":"32794bd1995d520f39a08b8a3ab60fdc53b049ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 10:31:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16643/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b9cbbbd693fa7289585418156a4b701636b68628","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:39:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ceb43a2bcff454894f519191e8c6feb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/733999f8d3904deea61000b7a3180f84 : FAILURE in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c37f910040d42ff9b6d56ae2ed344cf : FAILURE in 4m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/96df5912294b4e3db7add0026c66ed93 : FAILURE in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"323af3c39f1c89fb048b73c74ce1c5ba7b5c0a08","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 10:40:38.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e8eccf8b3a40400982232a7cc3f39d48\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0f71da02688342319ef0eb8378c2c157 : NODE_FAILURE Node request 200-0000002788 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"9b11267f8cfc65a103183d6879491e2f8431b3da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-15 11:03:27.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"72a119729dd9c3cff05b02c26e40818dbc851b94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 11:03:35.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16647/","accounts_in_message":[],"_revision_number":4},{"id":"1cbfb16f41858b2f264e78c67ff8f7f71b196562","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 11:03:47.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16647/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"4d5188a21fc52b15d33fc6a0c1ef0328ac8edd1c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 11:12:06.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5916ef2233f04f199e3eb4abb31d83f3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/31f51e87a45a452189046fa60d861184 : NODE_FAILURE Node request 200-0000002792 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"c9dfe89093d7b0ac1624f88298604df5f29bc022","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 11:19:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f1caf3e50534440a4e4c53e88c5c19c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1d81bb7bc894094a411ffff98ee05db : FAILURE in 14m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30d54e7c8a3b4054ac61349bec15dbdf : FAILURE in 11m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2d1a49769e174ddab10540a097908b41 : FAILURE in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd7c14d67349029154d5d63b379a816cc1cb5f87","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-15 12:29:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/54/.","accounts_in_message":[],"_revision_number":1},{"id":"54ace7ee9fc9291919939b0da2a494b05387a3c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-15 13:01:56.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"55e6b091762b30b4239c6a63de9bb79247b062e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 13:27:13.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a204f285574d4f02bc59bade942f5d40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96cdceb71aeb44b68437365e1ea0d4ec : FAILURE in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c855c34232d6453e81b84ebd85aaf498 : FAILURE in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a41cb9bbb3374d2695c8aac9dc7e66ad : FAILURE in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba855a1284b247f6865146642dbfc312 : FAILURE in 5m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/377a8e8847984b49a8c3fa215796847a : SUCCESS in 1h 00m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d8c16a89df4844b0b304b0afc5089822 : SUCCESS in 51m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed74b0bcd06c40418584ad085c405546 : SUCCESS in 1h 31m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c13ba0bcdd84fc0bd5a27fff20863ca : SUCCESS in 53m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adef7dd78a9b489384e4a9ce2f377b0b : SUCCESS in 14m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4bd8f1d9f3234dcf8c2bf6fc946854a8 : FAILURE in 5m 35s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0c065975606e4390930b352cb97cf6ce : FAILURE in 8m 16s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6ab31cec0a904f908515904954cc63ef : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b18f478cf444273b314e3073ab2de45 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d01dd574b8f6441592ef376d3ebddc96 : SUCCESS in 8m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b5daafb995794d5bb21ae3923c027318 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/14aca40d4ead48b7a3b2daf226f52676 : SUCCESS in 47m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f763f4e657c14b2988030e1f5708bf5a : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea35cd1a142b47d0a03341bce23788c9 : SUCCESS in 2h 17m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dd09fd5d1a314105a5072ef154830106 : FAILURE in 2h 05m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9bce6ce443c04996925a7a731af68552 : SUCCESS in 1h 43m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/674af4544b724c10b4bb5a6694dcc029 : SUCCESS in 1h 47m 57s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a537694931ef49658c0cee02c2e05a74 : SUCCESS in 1h 12m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/36a16e8075854e8bb4ec3e3ef39f0f30 : SUCCESS in 1h 36m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/47fc1410c2ca42dd813b45a874218125 : SUCCESS in 39m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/23dba0be5dbc489ba091b4a61b43de4e : SUCCESS in 1h 24m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/16c39984ffdc494ea06c9152748b2939 : FAILURE in 1h 17m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aaf4a55ce2204f228b956ec7d2dfe62a : FAILURE in 1h 31m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fac0851dc64e4dc0b438f355da02b713 : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":4},{"id":"1082d5287a9bf51afba0a7cf1d2fb33a48915c11","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-15 18:32:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 33m 02s","accounts_in_message":[],"_revision_number":4},{"id":"e7d193207bc45deaa0164189c6496ae27e3750de","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-15 19:02:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/60/912060/4/check/fujitsu-eternus-dx-iscsi/201b803 : FAILURE in 3h 02m 58s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/60/912060/4/check/fujitsu-eternus-dx-fc/c2d94a7 : FAILURE in 3h 20m 28s","accounts_in_message":[],"_revision_number":4},{"id":"7fe4285e993d225a6fce6847a3ed9a5b123960b7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-15 19:44:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/59/.","accounts_in_message":[],"_revision_number":4},{"id":"9bb3265cd27c92e854661692148c19e4df4c2d6a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-15 20:08:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 32m 38s","accounts_in_message":[],"_revision_number":4},{"id":"4ccafc6868a2e890c3dfc01557a76ed27695e538","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-15 23:01:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/912060/4/check/DellEMC_Unity_FC/c8e0316/DellEMC_Unity_FC/1517 : SUCCESS in 1h 14m 47s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/912060/4/check/DellEMC_Unity_iSCSI/01fa6df/DellEMC_Unity_iSCSI/1549 : SUCCESS in 1h 50m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"f23bbd33069d9ccf4e96c75700eb06ce4fd9d0c0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-16 16:26:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/912060/4/ : RETRY_LIMIT in 2m 34s","accounts_in_message":[],"_revision_number":4},{"id":"87a77ce6c5b2c8292aec92ab6a20d228371ff844","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-17 16:13:43.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/4/20240317133157/NEC-V-ISCSI : SUCCESS in 6h 46m 43s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/4/20240317133157/NEC-V-FC : FAILURE in 11h 41m 42s","accounts_in_message":[],"_revision_number":4},{"id":"810280687ffb71b4f3c42831afbd51363116b3c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-18 05:39:25.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e1d4f149721173be05778466bb338d315a18018b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-18 05:39:34.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16655/","accounts_in_message":[],"_revision_number":5},{"id":"6c8d7cc3fc84f12f7e01e573a1894823224f63bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-18 05:39:46.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16655/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"6e8f67418177bb454449225f413628b307da3b26","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-18 05:45:05.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/74388f2429c9464c81d41d22ba7051e9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b5d11c41c6504b408e7baba900de49b4 : NODE_FAILURE Node request 200-0000002802 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"223a8b3fb39449abeb0c26b825e16a3792b18c85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 05:47:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cd83648543247c695d8373afed0e174\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99ed36094ce44478694c00c73bdc0d1 : FAILURE in 4m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03d65d7e86014c19b0543d8d49431056 : FAILURE in 6m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a40d976963a4085ac73607d941ac533 : FAILURE in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"315dfe9e7e4cfd2dbe1d01249b27a3847af6245a","author":{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},"date":"2024-03-18 07:45:00.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"083b21b19e34bbb509d4a6e3024cf6a033e07cab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-18 07:47:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":5},{"id":"26d61f1e894dbd670e407f4e14cf717b6402fcda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-18 07:54:09.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"49f7e9b8ddd74bf6eced0b283fa7f20dd61f9aca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-18 07:54:18.000000000","message":"Patch Set 6:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16656/","accounts_in_message":[],"_revision_number":6},{"id":"1a9332316d0845792273116f97304fbf2c57acf6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-18 07:54:29.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16656/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"3e377ac79f29c68b7a40f4f91cdc95644ac36a24","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-18 07:55:33.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"43315c7c240a3f0971efe2606c1c43ccd790f7bc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-18 08:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/5/fc : SUCCESS in 0h 28m 26s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/5/iscsi : SUCCESS in 0h 43m 31s","accounts_in_message":[],"_revision_number":5},{"id":"f5e47a4ac1d5972a5995dda2cdd65bf0f1068413","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-18 08:00:08.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/77dba7742bfe4ffb9dec20bd79452a01\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a6bdd3b42b004028b7f928823c7b946f : NODE_FAILURE Node request 200-0000002803 failed in 0s","accounts_in_message":[],"_revision_number":6},{"id":"7ebadd1a62f9c1fa431df49e5a82d9f1d4d0ac75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 08:06:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/072ee996d1c745f8b88375f5562b834c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b0b6c18609840debdc2d8fb85058a60 : FAILURE in 5m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5a19409891742748284d3a52ea9492f : FAILURE in 5m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fcc7ae57babd46f9a5a33a2261232de6 : FAILURE in 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a9e1e26ca5969edbb156615cd388348226037f70","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-18 08:52:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":6},{"id":"adefa3db49e6f073ee0f8d759650461d0ff8a577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-18 09:01:03.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"63d6fb638a5e6a8f4a35af8035fae8588eb2c1fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-18 09:01:14.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16659/","accounts_in_message":[],"_revision_number":7},{"id":"8b711007e527d4dc09dbb1d2d2aa41376b578d8b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-18 09:01:25.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16659/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"8e9c1513837b4c84a3c5acc078550c6e79d1b54b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-18 09:07:21.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/215ad6c832334e88a04f225e0b67e835\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a9ba9de463444056a9920ffb59543e7f : NODE_FAILURE Node request 200-0000002808 failed in 0s","accounts_in_message":[],"_revision_number":7},{"id":"1e2de2a17e129750751553f333af82e2790fbe8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 09:10:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c9b7003506c475ebcce12ba2c0c2ddf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfc5982eaad14d8ca266c126a4d5f37e : FAILURE in 7m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4eaffd6720114d608631b72290763bce : FAILURE in 6m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/11b0599b5bc74ada8612463b86812834 : FAILURE in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9450111f362abb6f352b42c6e7f8a17ece3e8bbc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-18 10:00:06.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/6/fc : SUCCESS in 0h 28m 29s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/6/iscsi : SUCCESS in 0h 44m 18s","accounts_in_message":[],"_revision_number":6},{"id":"917f831a7018ed9d43e923dae636be800e56b907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 11:32:21.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/888d29ba0f41480eb8451f632c5fa753\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aac6a73d513487e8eb91f602ec71ecc : FAILURE in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/528be6546fad439b9a753546e059912a : FAILURE in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6790ebbe46874ccd89440872f928b48d : FAILURE in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d6f9e2a8a943abb6ee9973d2b839fa : FAILURE in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/c54f6a466b884115a34445a37d5f4f3f : SUCCESS in 53m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/06f336a771f74ef6acc159ec1bb976ed : SUCCESS in 51m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd9b4d7b6a90482da824a984cff7cd0e : SUCCESS in 1h 18m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfa09181da964109b528c14c5073b2e7 : SUCCESS in 53m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df6b85ca7d634fadb9afc4c883771239 : SUCCESS in 11m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c47c9e77e5724a5e964d37df27013f21 : FAILURE in 6m 19s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fb10c3c0009541ea99e0dcf95e3bd58f : FAILURE in 7m 20s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4124807a9ffb40aa97692e98bb8173f8 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9eb6b3015dc146269efd6a6dd12567b9 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19c8d43799344e85881337fbe9011d03 : SUCCESS in 7m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/540a3b15e68444feb72390e04e5ce9f5 : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9d41a783197045abbfe605f4e9573f1d : SUCCESS in 49m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db8d18771e2c413fa90b8547ea3b8ebb : FAILURE in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc88f84634524b69ab80a031dfc29702 : SUCCESS in 2h 23m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/02dd8ec6750d433cb108a580dccd01a2 : SUCCESS in 2h 01m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f18600a82a742e5a3e103974c7db9f7 : SUCCESS in 1h 26m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e83b8db12b1842ce9f12f52986dcb275 : SUCCESS in 2h 02m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6d25d7c8508f4bb684639bf866567588 : SUCCESS in 27m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b0a844c2594d4c7ba03f6c56d7e2126a : SUCCESS in 1h 14m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/04658aad449d4cd6be3edf735cf24bd2 : SUCCESS in 1h 30m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/affc02666ca34b6e9c4f32a084959b88 : SUCCESS in 32m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/575ece9f9e4d4d3b9c68cf4e565bcb5f : SUCCESS in 1h 16m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a5abc9894f0d445d82cc4c7ea192cd78 : SUCCESS in 1h 18m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/64105fe5d38c41d286490c37dc431d16 : SUCCESS in 1h 42m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01a35e58b41344ae8efd7b39dcdd2544 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":7},{"id":"253a9a08b9e7c875d82ed701ab5365e74f4ae7e3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-18 12:19:42.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":7},{"id":"42ea1d680b72646864baf8dfe831c03c1edf75f3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-18 12:24:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/60/912060/7/check/fujitsu-eternus-dx-iscsi/8a9c28f : FAILURE in 3h 07m 50s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/60/912060/7/check/fujitsu-eternus-dx-fc/887c902 : FAILURE in 3h 10m 38s","accounts_in_message":[],"_revision_number":7},{"id":"b09958874721b810bb8220e959fcbd97a8564812","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-18 14:17:08.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/912060/7/check/DellEMC_Unity_FC/8a7e960/DellEMC_Unity_FC/1531 : FAILURE in 1h 28m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/912060/7/check/DellEMC_Unity_iSCSI/b7849ef/DellEMC_Unity_iSCSI/1563 : SUCCESS in 1h 36m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"96e557cbec71e03591dca9eba6e34d5754d1a06f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-18 16:58:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 31m 45s","accounts_in_message":[],"_revision_number":7},{"id":"e672e083b91657a38a9b653b6658ffbc95b0a6ee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-18 17:25:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 22m 44s","accounts_in_message":[],"_revision_number":7},{"id":"350f1e31e923377b35637c6689ac9dfd2eb404ee","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-18 21:55:42.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/78/.","accounts_in_message":[],"_revision_number":7},{"id":"65b4830b94cb73148c5c4cabd2055e1c4800d059","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-19 01:11:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/912060/7/ : RETRY_LIMIT in 2m 34s","accounts_in_message":[],"_revision_number":7},{"id":"d159530fef70249a40260fd5ad7eb11f3e3d17de","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-19 10:33:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/7/20240319122510/NEC-V-ISCSI : SUCCESS in 3h 10m 23s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/7/20240319122510/NEC-V-FC : SUCCESS in 7h 08m 33s","accounts_in_message":[],"_revision_number":7},{"id":"459826cdc0516d113062c910c276b466cefe5953","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2024-03-20 10:32:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"58d3a1c86b99448a59f0e53700246c09f755ded5","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2024-03-21 06:47:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"51d4a36060799a1cb4677f7b0723cef217dd6a2c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-21 15:00:07.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/7/fc : SUCCESS in 1h 04m 28s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/7/iscsi : SUCCESS in 0h 55m 56s","accounts_in_message":[],"_revision_number":7},{"id":"7e852c381cd576fdd1641da1291f37ee9c3937d0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-21 20:22:13.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/7/20240321200917/NEC-ISCSI : SUCCESS in 8h 36m 51s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/7/20240321200917/NEC-FC : SUCCESS in 9h 12m 53s","accounts_in_message":[],"_revision_number":7},{"id":"11fc71ab8667aae738622f2c231d3dcd10d2c32d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-27 12:58:09.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"ebe07e5de2a55b4a2a957b65b6082d9c3147a694","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-27 13:02:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"26116180670ae8d0e70862d550e1a35d0be79ee6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-27 13:04:06.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a2bdd7b3d40a4c4096928867540c88d1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4358f4844db844c29b1cdc58c640efd0 : NODE_FAILURE Node request 200-0000002886 failed in 0s","accounts_in_message":[],"_revision_number":8},{"id":"74bb892f98517e72f042861b79695a4183a72d1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 13:04:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69f0df38fef44f3a1b3531395b2055b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/170d2265b4bb40619a276621a24df660 : FAILURE in 4m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c5f83450eb34a0ea7c47a4eb2eebc70 : FAILURE in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"147134263001a28050a4c205e8b16f40cbe4660b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-27 13:25:15.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 25m 23s","accounts_in_message":[],"_revision_number":8},{"id":"c216c337e5eaadce42d0b52fdf47a460e943e608","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-27 13:56:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":8},{"id":"0ae1dc78fd19540f9848240d713193bfb9b8bdc8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-27 14:03:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 34m 12s","accounts_in_message":[],"_revision_number":8},{"id":"68076a8381a4c2eb270395a144f3c5aed113af70","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-27 14:12:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4fdd4c46241a469d9180783da3734c0f : FAILURE in 1h 12m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9e984ad852c746f89f21121ec6b38228 : SUCCESS in 1h 10m 55s","accounts_in_message":[],"_revision_number":8},{"id":"aeceb91af49ed82b105688d925add364c0e26671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 15:02:01.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14f4af0c898b46bca64f54e3ede6a63a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57f170359fa43ac9ee716ccb1f0a87b : FAILURE in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32bc36d998ea40e39665f758fdbcb106 : FAILURE in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1fcf4e9a3e641dca65dcc0deebed709 : FAILURE in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/859f12d91df64b168469db01e8f2a3ec : FAILURE in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5cb031df154c1c9f53e59b316121b6 : FAILURE in 6m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/de5f80d300be407eb716127e070eef03 : SUCCESS in 1h 02m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1af66cf66bc3409a8080a2cdee74a2ef : SUCCESS in 59m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b9809002951442839a59c0723d013d40 : FAILURE in 1h 24m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/949103b9f2fe4857aeb2a7be8de4de24 : SUCCESS in 47m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96c4209faeaa4962973bea167f989328 : SUCCESS in 12m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cc466f0a09b043e7b923579316d25389 : FAILURE in 6m 09s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/88091be31bc649c484d5b4d275cc2744 : FAILURE in 5m 54s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1211efaa67ba4eaaae8b92d13cab10a3 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8e0762c8b2d409a8458794b69a23221 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e7c1eab044d435093aa67399e7435a3 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dee5b7c53dac4e2db0bba7f505cef396 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f51f7683179a4dca99395c13e494c75f : SUCCESS in 45m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b18a4246f7f54051915b919336338be1 : FAILURE in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8de5a40602a24240a2de1a2a5f191ee8 : SUCCESS in 1h 44m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b1762d20606e4da9a3b7bf3a284062c3 : FAILURE in 1h 46m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9f88e79dadf44a26ba6b91d855bde700 : SUCCESS in 1h 54m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/92e41b7f04644f36878544ad690f600b : SUCCESS in 1h 43m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/60ccdac22e884a728b651708ea4447cf : SUCCESS in 36m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aa6bf008603247e2ad6f728e5990244b : SUCCESS in 57m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5f33f76da8424291921f6cd5b1fe9e5c : SUCCESS in 1h 35m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8a2388d69a694388bc985f1f5f641735 : SUCCESS in 46m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d123bd366fe344ed8d95671326fbe002 : SUCCESS in 1h 12m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5cca1c60e2d745639a238778fffb2318 : SUCCESS in 1h 41m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f26ed256a4d844179a3cce04ad3aa644 : SUCCESS in 1h 56m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09a1b0bb63bb41f883e502b449f2e3c6 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":8},{"id":"373280e4f8cb04d659776803596f55cea92b0d07","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-27 15:34:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/60/912060/8/check/fujitsu-eternus-dx-iscsi/7bbdc57 : FAILURE in 2h 36m 00s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/60/912060/8/check/fujitsu-eternus-dx-fc/bc2d4ef : FAILURE in 2h 22m 39s","accounts_in_message":[],"_revision_number":8},{"id":"3b45f9f8e3b866bd935893796037e8dddec4bdf6","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-27 16:00:06.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/8/fc : SUCCESS in 0h 39m 24s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/8/iscsi : SUCCESS in 0h 52m 28s","accounts_in_message":[],"_revision_number":8},{"id":"e975ce338cea91e957768aaf130e6fb1fff590dd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-27 16:05:18.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/912060/8/check/DellEMC_Unity_FC/4edfef0/DellEMC_Unity_FC/1587 : SUCCESS in 1h 13m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/912060/8/check/DellEMC_Unity_iSCSI/91d9819/DellEMC_Unity_iSCSI/1619 : SUCCESS in 1h 26m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"6cb664072b9d53eafd5ffce47dad9410d460b96f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-27 18:37:52.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/912060/8/check/DellEMC_PowerFlex_3.6/a60e86b/DellEMC_PowerFlex_3.6/1098 : SUCCESS in 1h 58m 02s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/60/912060/8/check/DellEMC_PowerFlex_4.5/a02876f/DellEMC_PowerFlex_4.5/76 : FAILURE in 53m 48s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/60/912060/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/b381fe4/DellEMC_PowerFlex_4.x_NVMeTCP/12 : FAILURE in 2h 52m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"b1f9c64a68c9ef7ecdc63bcd9592c271b4c9ffbe","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-27 19:57:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/8/20240328001603/NEC-ISCSI : SUCCESS in 4h 16m 05s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/8/20240328001603/NEC-FC : SUCCESS in 4h 41m 15s","accounts_in_message":[],"_revision_number":8},{"id":"01245a906b1dcc135cf877eb28be52ea5550ef59","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-28 10:49:46.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/8/20240328163816/NEC-V-ISCSI : SUCCESS in 3h 04m 25s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/8/20240328163816/NEC-V-FC : SUCCESS in 3h 11m 27s","accounts_in_message":[],"_revision_number":8},{"id":"4f4d321b0bc1a3a84a5de5a8d30f22e9fa579b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-29 06:31:14.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"4f58e5080ea613e7e19055a36e84c3eca97bff48","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-29 06:41:43.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4498cc85b5974c9bb39330eee411ed1b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1363ad317a0a457794ea7b45d5d851ee : NODE_FAILURE Node request 200-0000002899 failed in 0s","accounts_in_message":[],"_revision_number":9},{"id":"88ef4127fec9fb7f770a36e0be04904a48a5c991","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 06:43:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6f9f047c6884e1681a6222cf1549531\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f6a3c8bb50f4c8999e11f95b164fd3b : FAILURE in 9m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15dd784a07634d4a9b3c7cc96268d07e : FAILURE in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d536ba0cdd01c2a05de85952e2ea9f18ec9d1e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-29 06:45:59.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"7a7219220a52826eae91e69072b8b1bea3b8db36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 06:53:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e99eebbf3c1e41c19132c415e6383ce0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f160b0199a7437e93a5c828230d2ee3 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9824c39f2c3c4da0a5cf1102261281be : FAILURE in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"84a3270a9a0f80be0a79a9d010643d55e42ce00f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-29 06:53:33.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/de6a4a252b1e448a94535ddf40a80a15\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c69e4a906e1e476da8442a1e139af9bd : NODE_FAILURE Node request 200-0000002900 failed in 0s","accounts_in_message":[],"_revision_number":10},{"id":"6c4c750c380e4b0bb874d725d8e74b9e4fa6ef7b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-29 07:34:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 44m 23s","accounts_in_message":[],"_revision_number":10},{"id":"54c67a8acde463b0d6fda846dc34b1c42e4f066c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-29 07:45:45.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : FAILURE in 54m 02s","accounts_in_message":[],"_revision_number":10},{"id":"a137625dbc716b2c424d56bfbf44f6cdbb40ecd8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-29 07:58:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 19m 34s","accounts_in_message":[],"_revision_number":10},{"id":"c8d6b09cc2b4a4d5519fe78c0ffd152ef83637dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-29 08:10:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/141/.","accounts_in_message":[],"_revision_number":10},{"id":"6d419ceeec9f8a9339b4e2614e04cd66db001073","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-29 08:25:25.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/912060/10/check/DellEMC_Unity_FC/1450c92/DellEMC_Unity_FC/1597 : FAILURE in 1h 23m 49s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/912060/10/check/DellEMC_Unity_iSCSI/44ab3d8/DellEMC_Unity_iSCSI/1629 : SUCCESS in 1h 39m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"03f74782c5edc6ae4093c95ca0e2e1b085f5b91a","author":{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},"date":"2024-03-29 08:47:53.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"de496017c94338a59c76289880f9965cd68fc397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 08:58:34.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4903cacfc9cb4ade93a0c280f8ebb022\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3736b96f28414fbc96532e96e8f0b2d1 : FAILURE in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7abf35ed6744c6d919ff46ac2ad0ebe : FAILURE in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fe64b64cebf4c0a93967fa1ce6530fc : FAILURE in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c468a69d64da4a2a9955f70f7d83daa6 : FAILURE in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d08f39abd8a436e868360a555cefde8 : FAILURE in 5m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/72f3f8a8d3d9487bb4f18262f7821e85 : SUCCESS in 1h 05m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b85a1fd822c6437cb2949cc4b395082b : SUCCESS in 1h 31m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b4af63e8d9f4f8dadb82d46269d3ef4 : SUCCESS in 59m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/263dd9a4dfc84b51a335bb93b07c5c55 : SUCCESS in 9m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bb9eb9e33a6344e2b3181ea6225aca91 : FAILURE in 4m 30s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/99e82044a06a44d3af8a32adebee4978 : FAILURE in 8m 01s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f9acd51040a84ae5bc2b7a0482428496 : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf3a3390d2424f54be2ca7c0cc0449ea : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f5adaad0dc0445a8ad8509bd496d0599 : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/209b7a23c0ba4825bcae488de9978cb2 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0e98a38616ff4daaad8a176684a2a6f7 : SUCCESS in 31m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2508994ba3b4890b13a42a8fea4281a : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/85575d5d6d374ad7ab476d5a6d788251 : SUCCESS in 1h 40m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c0b406bd76254034ac02c9c9c3f41343 : SUCCESS in 1h 52m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3cfe2ada84034768b11350efc997854c : SUCCESS in 2h 04m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6ad5eed11c024ef5bc673cfbe0db99f7 : SUCCESS in 2h 07m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/37a4734647df4ed3bc0162d4b330f7d9 : SUCCESS in 39m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e61168b3b2784a01a312798f67635b8c : SUCCESS in 56m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58a0fe24b8c047b59deaa179e5b8ad0c : SUCCESS in 1h 40m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0eaa03e211de4736867a6d37a30a8d6d : SUCCESS in 42m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/62c8c67171594b4d902d464f600e5078 : SUCCESS in 1h 32m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8694f37613104fb4b3a4c77c6771192c : SUCCESS in 1h 10m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4e7afc48aab4caa87b92b5d9a3ed624 : SUCCESS in 1h 30m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/425850116ec34bf99496f80a6df96996 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":10},{"id":"eb39056a4ebe50b5f038d45c99252cdeb5e7160e","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-29 09:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/10/fc : SUCCESS in 0h 40m 05s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/10/iscsi : SUCCESS in 0h 53m 49s","accounts_in_message":[],"_revision_number":10},{"id":"fc7b11ba9147f6a3922a339ece4e58bf4dadd634","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-29 09:22:39.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"6d24b0f875c2811763b15757f6e1c3092a221371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-29 09:25:48.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"6d5dffa258d3f1d118bf9c12e8f8762bbaabdd84","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-29 09:26:30.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"1564d15f7195ccbd94cff4b14e5a2817c8f12afe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 09:39:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd30579c22334267addb30191a339409\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb00e394f4314f43a69b06bc1f865d97 : FAILURE in 9m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3ca55058065c41bf9210d3e37a4ad1e1 : FAILURE in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"320b5567b0de14d9d3c18c8fa03466be9e23d2a5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-29 09:45:38.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/69610b975bcd4129861370520a5091fc\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/05b45d395e444ff2bd9cd910cdf45b27 : NODE_FAILURE Node request 200-0000002903 failed in 0s","accounts_in_message":[],"_revision_number":12},{"id":"e40fda5a5c0051de89462412c290f3417fc8ca66","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-29 10:00:06.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/9/fc : SUCCESS in 0h 44m 32s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/9/iscsi : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":9},{"id":"7cb76eeb94d01d51dad4a6b52eb03d0ead9adcb2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-29 10:29:05.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":12},{"id":"daf2531752f070ab17074f42eab1b4e074cf2a5c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-29 10:35:14.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 56m 56s","accounts_in_message":[],"_revision_number":12},{"id":"12d4d3946e0b4ddf785024428812ea2f9de11783","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-29 11:05:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/143/.","accounts_in_message":[],"_revision_number":12},{"id":"06e16fed978e6367c8b8ab5188ff99c71338b593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-03-29 11:14:41.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"0ef9e4059ea833058d504ca2fd13d5c410a21234","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 11:22:08.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76587d45f01e4b949a4658dbae086189\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9d458d59efa4852a372354a3410444f : FAILURE in 5m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d744a673dd284fe1bcfd467086bf5f38 : FAILURE in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8d18bcb7fa382897247b7a78bd32f6637f355468","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-29 11:25:40.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d90422f35906424182cba3efe36657d7\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7b65f4bbb8e84aa4aef73e53d7d8c6ab : NODE_FAILURE Node request 200-0000002904 failed in 0s","accounts_in_message":[],"_revision_number":13},{"id":"8fe7a3197002e0844976c9373e5637a2ddd93a8b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-29 12:11:42.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":13},{"id":"f22073d6b498c9b7f756abf38c32333544b871b5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-29 12:14:59.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 55m 34s","accounts_in_message":[],"_revision_number":13},{"id":"484e607d27876cb2ae0c0b3233f7ea3317953d6c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-29 12:43:57.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/144/.","accounts_in_message":[],"_revision_number":13},{"id":"17ce1b6fa8bc08dda390dd3f8c6427020b71e0d3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-29 12:56:57.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/912060/13/check/DellEMC_Unity_FC/7aa9995/DellEMC_Unity_FC/1601 : FAILURE in 1h 23m 48s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/912060/13/check/DellEMC_Unity_iSCSI/ba629e5/DellEMC_Unity_iSCSI/1632 : SUCCESS in 1h 41m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"de98f6ce7931a87884db702f3142187d61a031af","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-29 13:00:07.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/12/fc : SUCCESS in 0h 53m 32s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/12/iscsi : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":12},{"id":"fd9a5e1ff181bc85d2dea31eda8d43f164d84aa6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-29 13:17:57.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 58m 25s","accounts_in_message":[],"_revision_number":13},{"id":"1c229c2c556bbad9bbbf83542b61a7e74f1bd71f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-29 13:20:42.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/912060/13/check/DellEMC_PowerFlex_3.6/c2d6d09/DellEMC_PowerFlex_3.6/1107 : SUCCESS in 1h 48m 58s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/60/912060/13/check/DellEMC_PowerFlex_4.5/30f3c11/DellEMC_PowerFlex_4.5/84 : SUCCESS in 2h 05m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"73ac8037d246d4050e8e1be351636e181f36304a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 13:34:50.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0439d5d52a6647d6822b6a7425bd8ac2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bfb1d5032fc467b97c70c662b05ed65 : FAILURE in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/774fbe77afdf4cd395bb26bc4bab776c : FAILURE in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47a95140d17b4a93861850cf3b4e9e08 : FAILURE in 6m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1924f7f71a6e46f19ffc15b7cefeed25 : FAILURE in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ac8b35404545bca98d38175d7407a8 : FAILURE in 6m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/b1e94ed1d4f940bc804d8812a750d9fa : SUCCESS in 53m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ced0e6f2ab44655b87b1cbc124dbf62 : SUCCESS in 1h 26m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1567167bfc7f46219e346e58b2a24f0c : SUCCESS in 1h 06m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87f64aab144848f194fa21e805ca7656 : SUCCESS in 11m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f998c665b8a4460b9448ecbab2199dae : FAILURE in 5m 01s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1d8be87de211480c84eac78a665005b4 : FAILURE in 7m 07s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/63f948c64fae4c78b4881db8e28e77df : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b18e43a66ff444c98ce43a3f7670cfb : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/abfc34f32a634a9db831e2f363975adb : SUCCESS in 5m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a304950cdbe14ea9a1919c6a31332b6b : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea96ce0ef4894017ac456583d81b46a5 : SUCCESS in 49m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f581b8714d4747b39f4730cf68abf846 : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ae9acd99eebf4848b5d0a3a1ee351e29 : SUCCESS in 1h 58m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c19c63a1bacb4168aa3b3193cfbfb990 : FAILURE in 1h 59m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/915481c9045740c7afa76ffd2d167826 : SUCCESS in 1h 54m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1b46c17695594a72b95df361bd49af16 : SUCCESS in 1h 59m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aba7f992ca1a44de9fb15943af175956 : SUCCESS in 38m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ef69865e5234ca9a9b680232e166334 : SUCCESS in 1h 22m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/22fb3811964c46e3807338e1a4f94201 : SUCCESS in 1h 31m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f25b75f52d3e438b981611bd49100e3b : SUCCESS in 43m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d1ed8dc4c9f4407887ae8fbc93e5b40d : SUCCESS in 1h 21m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7e155acb19a848359336e2e593b7f5f8 : SUCCESS in 1h 16m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4afa6efff14b4ad79d51ca358cf50f19 : SUCCESS in 1h 12m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25a59c50211647d4bcd3ae9079dcd257 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":13},{"id":"364d273b641e27efec2499c2ea7a986db923646f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-29 13:43:37.000000000","message":"Patch Set 13:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/60/912060/13/check/fujitsu-eternus-dx-iscsi/7bce11e : FAILURE in 2h 23m 29s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/60/912060/13/check/fujitsu-eternus-dx-fc/6dc1aa5 : FAILURE in 2h 27m 15s","accounts_in_message":[],"_revision_number":13},{"id":"42d2b0d4d2e978565eb99c55f7135714e9e96bab","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-29 15:00:08.000000000","message":"Patch Set 13:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/13/fc : SUCCESS in 0h 42m 20s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/13/iscsi : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":13},{"id":"a1b0540496ce6494004b6a19f302d3c50ae956eb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-30 09:59:51.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/13/20240330141926/NEC-ISCSI : SUCCESS in 4h 16m 49s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/13/20240330141926/NEC-FC : SUCCESS in 4h 40m 21s","accounts_in_message":[],"_revision_number":13},{"id":"bb5bcf2c68e76ddf13e42275da0e5d02a49fb9b9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-30 11:44:08.000000000","message":"Patch Set 13:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/13/20240330084234/NEC-V-ISCSI : SUCCESS in 6h 34m 11s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/13/20240330084234/NEC-V-FC : FAILURE in 12h 01m 31s","accounts_in_message":[],"_revision_number":13},{"id":"5b75ca43cd1a602869ed7587f3dcae806e42d9a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-04-01 06:58:00.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"e3f9bcd4c00a568f56870a883b9e634b9e0d1514","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-01 07:04:21.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4a33709fe59b4e2cbc974dc8ef669bba\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/393030f3051b46d9bc463ce73b4728aa : NODE_FAILURE Node request 200-0000002913 failed in 0s","accounts_in_message":[],"_revision_number":14},{"id":"ccdd30879ca8e042276620b74f4f0ba0349cfb11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 07:10:28.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7d84acc688b4c8aaa31e5a44fdf1710\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e0a4b3ff77e42d3bf5adc44ba0ca43a : FAILURE in 9m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a299b1467ed54f9fbf3cf44f344f0db8 : FAILURE in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5b69f832e5b2cf0e976cb32506a43e708bdb967d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-01 07:38:51.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : FAILURE in 17m 25s","accounts_in_message":[],"_revision_number":14},{"id":"dfda1ad85680cd7f792c11e3bdbfd31ed1ebbd0e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-01 09:30:48.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 50m 00s","accounts_in_message":[],"_revision_number":14},{"id":"eee6648bf7943cc5087e878af9b894894e14d1f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 09:46:29.000000000","message":"Patch Set 14: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d84fb65be5ae47078c89b6c10b351c1a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f36fa314c147d4b725625e08a3dd2b : FAILURE in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0e2675db50049d99b93467a7a8af7b0 : FAILURE in 8m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/538259fedba548538680df20f64efb06 : FAILURE in 9m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbdf85e6a0a3480fb4cf5392c67b15d0 : FAILURE in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919d1d852b564012ba7750d393465dc0 : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/410f211253124b448b4674c94508fe5e : SUCCESS in 41m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f55e1bf33b7042d28fa268ef1559ef5d : SUCCESS in 1h 41m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9954c1cc1d254a0ca885bb321396471a : SUCCESS in 40m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04291ba02d064ef4aedb79f7eef38bae : SUCCESS in 11m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3b849a16e04744e5a2e9ebbff5609301 : FAILURE in 11m 53s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4e477c2ab76f4b749fbf2cef10bdd50c : FAILURE in 12m 06s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6e2140a0c8a848fbb29cad1ed0bc6729 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04547917f1584697a08bb1793ba2c0a1 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f9ae867853424848b864eacc666ed129 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f59ab144b9f4407b0b4726d15b93d80 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2ad4fc66429f4bdead807083c8652583 : SUCCESS in 36m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2384e218acd448e1a7ded9d3bbefb0e3 : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/30b16ceaee8d459691aa051ba9896304 : SUCCESS in 1h 49m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fe2f3b1e6ece4d6da140e2f75e1374bd : SUCCESS in 2h 24m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a60000d51da44f9fb4ccdc414e33429c : SUCCESS in 1h 40m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8753730249654afc81d99431cde3683d : SUCCESS in 1h 51m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/83b8bd12a331478d9a098cc191ebd7e3 : SUCCESS in 29m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7a1c46178aa64d219599a7cbaa8a35ae : SUCCESS in 1h 08m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f4488b72e9884d58a9a4fc449fafa8b7 : SUCCESS in 1h 31m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/18ba5e0bed3f4ca8a0f89d6da6772997 : SUCCESS in 28m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c57ee37684f44e12afd6de1589838945 : SUCCESS in 1h 24m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/670469872fb1417da44a924157bdf0e8 : SUCCESS in 58m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8ab84ee6f43f4b06bafb21ff343fd61c : SUCCESS in 1h 28m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/112d71413796454a8de3aefbec1be8aa : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":14},{"id":"0b815d875992043ca2c51ef608f98ea9b8141224","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-01 09:53:04.000000000","message":"Patch Set 14:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/154/.","accounts_in_message":[],"_revision_number":14},{"id":"bad5aea7ae127df71a5c1d5fa88c20811d5e55cc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-01 10:11:58.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/912060/14/check/DellEMC_Unity_FC/1e1ff8a/DellEMC_Unity_FC/1611 : FAILURE in 1h 16m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/912060/14/check/DellEMC_Unity_iSCSI/e97fb6b/DellEMC_Unity_iSCSI/1642 : SUCCESS in 1h 43m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"e8c468457e2a0fb59a0580e03b95ec029f21d2d2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-01 10:15:23.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 40m 29s","accounts_in_message":[],"_revision_number":14},{"id":"638dad025bbcf69108635e145e585c811ac1b10f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-01 11:00:07.000000000","message":"Patch Set 14:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/14/fc : SUCCESS in 0h 58m 26s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/912060/14/iscsi : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":14},{"id":"52e87d4190fd03790e5d9c292cebd4e26a3f4fa5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-01 12:00:58.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/912060/14/check/DellEMC_PowerFlex_3.6/3fd3e88/DellEMC_PowerFlex_3.6/1117 : SUCCESS in 1h 35m 47s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/60/912060/14/check/DellEMC_PowerFlex_4.5/953c67d/DellEMC_PowerFlex_4.5/94 : SUCCESS in 2h 21m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"e6996bcea8fd9fb747af461e6e3f200140709cb3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-04-01 16:17:59.000000000","message":"Patch Set 14:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/14/20240401211926/NEC-V-ISCSI : FAILURE in 2h 53m 06s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/14/20240401211926/NEC-V-FC : FAILURE in 3h 58m 30s","accounts_in_message":[],"_revision_number":14},{"id":"a3759dcfd3dd594117c2c7abba3cdd4aa07a1b1f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-01 23:09:34.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/14/20240402032929/NEC-ISCSI : SUCCESS in 4h 14m 47s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/912060/14/20240402032929/NEC-FC : SUCCESS in 4h 40m 02s","accounts_in_message":[],"_revision_number":14},{"id":"cf1a6ca8220c0c31e8b0e3ffbe3cd268763c59fe","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-04-02 12:54:02.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"33d52481104b5d2bef4704848537ac52367d3f66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-09-18 14:03:23.000000000","message":"Patch Set 15: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"64f7d50464e83bb32bdfa35965573489bcd3edc4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:03:32.000000000","message":"Patch Set 15:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/577/","accounts_in_message":[],"_revision_number":15},{"id":"11ef3ce871ea8cb985b59d34f6473595950c4490","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:03:34.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/15/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/15/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":15},{"id":"642853ed86733a8f7792f1dc3588c9c8e98b5412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-09-18 14:03:52.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"83ffbfe3c625dc69c653c8577554cdebe535e845","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:04:02.000000000","message":"Patch Set 16:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/578/","accounts_in_message":[],"_revision_number":16},{"id":"940b0371a2ce738eaaaffe650062a281e2b66579","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:04:04.000000000","message":"Patch Set 16:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/16/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/16/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":16},{"id":"fb88c21e401ea89c59faaf4e3f615b20cd1d3ace","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 14:39:22.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0abd4d20efff415ba1bbca4e39415ca1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5fb0771b5dd437bb094acaf4bda4fd2 : FAILURE in 23m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bdcdf9411d39437899968855f850408b : FAILURE in 19m 40s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"765c7433e1c80ec348c94692fd23bb442c852da7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-18 15:46:09.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912060 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":16},{"id":"516b5786037e5ca9cd6e1b8a34c4bb18fd50cbf4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 15:46:52.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a8d3eeaa6cbe4f60bd7ee27202aa2ef3\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2be7d6550b17419593e8806db05ac29c : FAILURE in 47m 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/648caeddee9c421a922222260182492d : FAILURE in 1h 32m 36s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b8099e49f2a94ebfb74dc978982e3592 : FAILURE in 1h 42m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f30a9391148d48c98335455818eee514 : SUCCESS in 39m 16s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":16},{"id":"f83cf43d61210ade2ceb62d251f7b708371e7354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:20:48.000000000","message":"Patch Set 16: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86beddf8fe6643c38154bca3a0398f29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd124183cc0843479508bbe0ce1be8df : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ec510b604dd420f8b20942585f03d3c : FAILURE in 9m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/615da5e9ebb941e2af9392aef18189e5 : FAILURE in 9m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3af7de1aecb485daac4d42389c68f64 : FAILURE in 12m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364e637c2b2f44f9b16b378a6ca3b320 : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/8b4560bbaf84453b97eff21d8b18f411 : SUCCESS in 1h 03m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/718196d9c60e4efbbef630ea97fdffd9 : SUCCESS in 1h 40m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e089f7de4b9436ab3c2e1b2689568f9 : SUCCESS in 26m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bbf33cd7fd345488e61271c46b89d6f : SUCCESS in 19m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7b6dbb165a1a465493e024cd0df51266 : FAILURE in 14m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ed176080b7fc42fa9ddf0cc8f595974f : SUCCESS in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1312a4b49539449e9a3b0fd43290cc61 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2bb18c6ab5a4b008564127693d854f4 : SUCCESS in 7m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/26db0dfb253c4a08b1915a90340eebcc : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a6694cd87b64d6588b8e3db763318ec : FAILURE in 40m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d0d635281091451d88929aa572f2b30d : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e927cf1e39dc48d59a5d06a4abd9fd87 : SUCCESS in 1h 52m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bb58c2dee8b74f5b9971b9b898ca4781 : SUCCESS in 2h 14m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c9d9b7dfc9a1418982ecde765302f04c : SUCCESS in 1h 49m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e88db88656394c98a057d6df0ee97318 : SUCCESS in 2h 03m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ea2bd331577d4b6d91a2f909a8083f15 : SUCCESS in 38m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8a49c8ec1bc14c8281844cb263f0434d : SUCCESS in 1h 06m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/54b585d18163460a8e85a0dd5082786b : SUCCESS in 1h 37m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/516ceca05af640e9b762ea49cd7b4dd8 : SUCCESS in 32m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7da8401094744e1d9b114192fc60db91 : SUCCESS in 1h 28m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fdac5006c19b4b32af6368c6ed5e030d : SUCCESS in 1h 39m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/972d699e1281409a8a0947203c260471 : FAILURE in 1h 48m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/546a2d8be75d4dbc9bc01f4edd27d131 : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":16},{"id":"8de80e4185b2ae0bc059fcde40fde5b0f4d8ed16","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-18 16:30:02.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : FAILURE in 1h 07m 17s","accounts_in_message":[],"_revision_number":16},{"id":"ff537a406160626f208780391d12e6aec025222e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-18 16:40:10.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 49m 39s","accounts_in_message":[],"_revision_number":16},{"id":"3d9a8d4e962e5d807c4135c69336487ff389f5d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-18 18:09:00.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dc3456fc76cf4f57a18e2079ff62e363 : FAILURE in 1h 16m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f4f4ce2a6f08490ca2a0ab5d0e468e8b : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":16},{"id":"640c302b203b41793ae3df008df60525a7b7291b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 01:57:58.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/912060/16/check/DellEMC_Unity_FC/bf90416/DellEMC_Unity_FC/2620 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/912060/16/check/DellEMC_Unity_iSCSI/d965468/DellEMC_Unity_iSCSI/2629 : SUCCESS in 2h 12m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"72cf450579d92faea4c39b8671e2b2416fb14d9e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-19 07:46:13.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/16/20240919124133/NEC-V-ISCSI : SUCCESS in 3h 37m 57s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/16/20240919124133/NEC-V-FC : SUCCESS in 4h 04m 36s","accounts_in_message":[],"_revision_number":16},{"id":"2cc2b450562ab45041c3feffc1d0ff357e37b83b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-19 12:54:42.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/912060/16/check/DellEMC_PowerFlex_3.6/2513c4b/DellEMC_PowerFlex_3.6/2114 : SUCCESS in 2h 24m 32s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/60/912060/16/check/DellEMC_PowerFlex_4.x/ffd748e/DellEMC_PowerFlex_4.x/998 : SUCCESS in 2h 11m 01s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/60/912060/16/check/DellEMC_PowerFlex_4.x_NVMeTCP/638ad7d/DellEMC_PowerFlex_4.x_NVMeTCP/938 : FAILURE in 5h 53m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"be085fc8ab1e8f1ed6c0dd4ff7115844ff0edee8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-20 11:22:14.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/16/20240920115150/NEC-ISCSI : SUCCESS in 5h 21m 47s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/16/20240920115150/NEC-FC : SUCCESS in 8h 30m 20s","accounts_in_message":[],"_revision_number":16},{"id":"b04f263f8cdbc7943cf8dc97a5dfbb71c06f5495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-12-26 08:49:26.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"d16a4958b7229712af88e6643dc0bcb96375b44d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-26 08:49:36.000000000","message":"Patch Set 17:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1127/","accounts_in_message":[],"_revision_number":17},{"id":"564ca763e53b09427811c925114ad119c4e9bcec","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2024-12-26 08:49:36.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"b5f60d48f9893fb2a87bac0811bf7773fa5aa4f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-26 08:50:29.000000000","message":"Patch Set 17:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/17/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/17/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":17},{"id":"d1630d1973a955903721f9e742779a9ae38b5080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 09:10:34.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909038e58ddf406d999235fe05cb221e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91cfd2896f0d47cf9d7f5d8614db7eb0 : FAILURE in 18m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/15d4c3df5c5043cb899611330812a55d : FAILURE in 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"21b8d43c3f448adc410ca8d975963ee5b4c21beb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-26 09:10:58.000000000","message":"Patch Set 17:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/167/.","accounts_in_message":[],"_revision_number":17},{"id":"42b491c1729617b6e25b85ad6eddf515122525b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-26 09:13:02.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/156/.","accounts_in_message":[],"_revision_number":17},{"id":"c502b00fc72ed4e609cde3f5db235a582811f602","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-12-26 09:32:43.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/912060 : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":17},{"id":"35db47115b1d71e4b73d1b0b746aa5656532ce96","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-12-26 09:33:10.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":17},{"id":"b67ce539a614b9ea98732bc714578115d36b7733","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-12-26 10:09:18.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 18m 18s","accounts_in_message":[],"_revision_number":17},{"id":"fcd36c5889ce71e1848db3abc5ccdbaea4a886fb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-26 10:16:21.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d3108895390e4eca85493f8e3e32de5d : SUCCESS in 1h 16m 18s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1427a8d154d844fdbf3d6e4255abd3d9 : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":17},{"id":"2d0e3e232f1eafaad17448314cebb0a3da4c0f39","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-26 10:18:02.000000000","message":"Patch Set 17:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7bd01ac784114e19821e45fccf0663e7\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39e566fcdb2745d1980e97736cf0c0c7 : FAILURE in 41m 58s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8f0e8130df734d0eb573e5704c790b3d : SUCCESS in 39m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/65248be3359340ae8ecc00cd3b3a0c3c : FAILURE in 1h 28m 04s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7aea8199768e49eb8df78ed15429546d : FAILURE in 35m 36s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb44da51a21c453d802f5ef096aa5b02 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b59ff1dd66d4985a7c14b2362364e4e : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b3b32d863da547c1b061234ffb996fc1 : RETRY_LIMIT in 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/14ae256d97e44525b22dfec4e53040a2 : RETRY_LIMIT in 47s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":17},{"id":"7f911fa6f377223484a85df7270ebfb8659ec88d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 11:56:06.000000000","message":"Patch Set 17: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6b3b99cd0934754b4ae9cafaf7d0e78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65db8ceacfee4fbfaec076932a0c8259 : FAILURE in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6653eb180db4b68b778a1e2e877b5aa : FAILURE in 11m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a758d2a4f4cb4960b88a43741cb517cd : FAILURE in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3879ffe1db9a486c821e2cfbc2210e5a : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/de3adf9cbfa94cd59bf7518b12105967 : SUCCESS in 1h 13m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/165b545cfc684c5c9f04c9cf9cb4c93c : SUCCESS in 42m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed2748803e824bfbb9adca79ec89947a : SUCCESS in 1h 15m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/220b8e46af464a9694fa1aa40ba886f5 : SUCCESS in 1h 05m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95e04b35611c4e0cb7b18af8321b0b53 : SUCCESS in 14m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/26586483786e4b3ab133468d9326fc47 : TIMED_OUT in 41m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e382fffe205f4b1ab59cae5e92f0f409 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f64a9862e3684e26bcb8326935d1647d : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/18ae2887dfaa40ce9e6e0bfd1b72f4aa : SUCCESS in 6m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/34f0463c2ef8474cad911e3a3a782f38 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb96333d51604e3fb41c58b934e04c3e : FAILURE in 22m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f607962b39144b6bbd38bc09d111a1c8 : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f3ce011dfc2649fd9b65f99f828c4c63 : FAILURE in 2h 08m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4d5d5753a0174dfd8c6daac3c5d0d63a : FAILURE in 2h 53m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d5678bc373a48a48a7dba7de10683ed : SUCCESS in 1h 41m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ea2b2990d57646748941a3071368d543 : SUCCESS in 1h 36m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d3fe3ae48f6d46369493132afb7a2c91 : SUCCESS in 28m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dd416710a71441b697824fc8d897cfba : SUCCESS in 1h 16m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/44da29dfd64b4b1ea3b9f399e92020e4 : FAILURE in 1h 51m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e0bdc950697e4455a65e4293e1ee61ad : SUCCESS in 29m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d1d5ac8f2a0349ba9325d51f6f665346 : SUCCESS in 1h 25m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4ca5c9fd3e7e4307bf9118b47d4c588b : SUCCESS in 1h 39m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/681baeff78444d3bbabf613bcc609ff8 : SUCCESS in 1h 49m 43s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/828fc86b02ae4b03abb9cd807253c908 : SUCCESS in 1h 30m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0689d2d96674bdcb66fe06433d61011 : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":17},{"id":"214b46904cf0432775f7e55b3ef769666daf5474","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-12-27 14:46:26.000000000","message":"Patch Set 17:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/17/20241227222259/NEC-V-ISCSI : FAILURE in 48m 35s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/17/20241227222259/NEC-V-FC : FAILURE in 1h 23m 24s","accounts_in_message":[],"_revision_number":17},{"id":"003708d475283b34ce1fd573905fa4fc39b50fc8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-10 13:51:35.000000000","message":"Patch Set 17:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/17/20250210212143/NEC-ISCSI : FAILURE in 52m 20s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/17/20250210212143/NEC-FC : FAILURE in 1h 29m 48s","accounts_in_message":[],"_revision_number":17},{"id":"9963d54a5de76fdc86e6048aaf3a5fa454a03f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-05-14 15:11:07.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"430f7cbffd1f19911ca82922dec832e73c129f03","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-05-14 15:12:15.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"b540b9f34a354f3df3e9ced35ec371f0291096f7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-14 15:38:19.000000000","message":"Patch Set 18:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/730/.","accounts_in_message":[],"_revision_number":18},{"id":"f48271560163a3f3802014113c499547880255e1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-14 15:39:17.000000000","message":"Patch Set 18:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/852/.","accounts_in_message":[],"_revision_number":18},{"id":"6f84d11f818c4581a8f75c0e28224c7198e810f0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-14 16:25:02.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":18},{"id":"f59231fb10da2cad6ea384a17027b376f54dbcc2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-14 16:59:20.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 23m 30s","accounts_in_message":[],"_revision_number":18},{"id":"e8d5d28a58845180752db479ff428502db75b6d5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-14 17:07:54.000000000","message":"Patch Set 18:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/36345e1dfbe74e74899ba75d7f65c0c1\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfab26de2cf149d08dd5930447c4dd72 : SUCCESS in 44m 14s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e20ab9429e714af088dd4840a9863f2a : SUCCESS in 43m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/94612277922244919b19763b0d7d9388 : FAILURE in 1h 56m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/87ee42aa2953427c9fe65ae8df087bad : SUCCESS in 45m 44s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ec7b8c5e60b4706988b95c2b7fc0d65 : SUCCESS in 56m 23s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a2a5a00a1fa24c84b4882398df14e227 : NODE_FAILURE Node request 200-0000051733 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd6c93e423324f279a444b9495772bd0 : FAILURE in 1h 33m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bcc15e596fe545b89fcf1d5305e4d576 : SUCCESS in 51m 10s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#68\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#416\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#82\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#416\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matc... (truncated)","accounts_in_message":[],"_revision_number":18},{"id":"48ed1b16a3e396f3d16e7557897bdc4dcf22f130","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-14 17:14:50.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6cbcb7827f7d4a6f98c8160dd9bef009 : SUCCESS in 1h 22m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/de88e55657ee4ffcbad54accb26cef8d : SUCCESS in 1h 17m 55s","accounts_in_message":[],"_revision_number":18},{"id":"1a526ef714ad90a9e1e963c52e84015d72ed94fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 17:43:49.000000000","message":"Patch Set 18: Verified-1\n\n(16 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ce9a479e9824042a4a3f640e2840104\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c598ffa11a7e491d91a927a63120c094 : FAILURE in 7m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c689e15af564483b466475b44f15e54 : FAILURE in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/209ca42f0421458b9deb45264ac40bce : FAILURE in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbf0998dc13147c484c2e4137c84fcbf : FAILURE in 16m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e057217ee6b4742a8f032f9521aded4 : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/00a91a3faa1041f1b34d71f139b059c9 : SUCCESS in 57m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b70470cee31f4b4e8e24280eefabd70a : SUCCESS in 32m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/37b2567202024cde967e8560e0da8e92 : SUCCESS in 1h 24m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6748ccf62b0246b3a5eac693880b8fee : SUCCESS in 32m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/526f227b4796479ab4fdae11ef3674b6 : SUCCESS in 18m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4288589fbfb64cf290ae6d38ec469fb0 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/903790cf032346b788b7ad1a9f858500 : SUCCESS in 3m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dc9d280b252147ce835a01876e1c0081 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7dd0a0fc565541a5a545baa4479b1498 : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ac5e0dca0f5a489ab5b6c8a22ff9de39 : SUCCESS in 5m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c4f487da658f47d89672b8d3be5cda5c : FAILURE in 32m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6634e8f5795d41f9a0633b674524bda4 : FAILURE in 4m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a6a2cceeb97a4cc596b97cf1c243c9cf : FAILURE in 1h 52m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4be5c0fdd55a45d597b9f3e37fba9e37 : FAILURE in 2h 26m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/286fd83fb50644fca5aec95d4dec322f : SUCCESS in 1h 37m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6cd46100af594ac7b265eee918ff3e6c : FAILURE in 23m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0286db21e7314564b597d43e6ce40b65 : SUCCESS in 40m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/309ad528dc0447819f542b62652e7c1b : SUCCESS in 1h 06m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d4f2de4a04eb46db8aa27999c9b0ab0c : FAILURE in 1h 41m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bbc852561f144baab89450ee4a0d7748 : SUCCESS in 28m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8a4fb9ebe94440c98b90de9201c0866a : SUCCESS in 1h 22m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5f6f572670da4f9c81e085924869bed0 : FAILURE in 23m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38c804065c2e40d9a841b551838e07c6 : SUCCESS in 1h 40m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/741f86ee8439474b80564a3db2ace106 : SUCCESS in 1h 33m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11125e5716d544bf910324d1be5e2fb1 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":18},{"id":"1a5ed227466ee2199d2a995e286246c0f2c21c61","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-14 17:46:23.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"71b699e1193cef21876b53397915cfa9712c004d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 18:10:19.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82dedaaece354b4fa694a9b4c19ab3c7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de867bda4a4e4079a2c044ed853ab099 : FAILURE in 19m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5bb78105155c4e879c88187a8a467e5c : FAILURE in 20m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0843ac95e6394b68aad0628b95823070 : FAILURE in 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"08b598f5bcdfbb6ee2fda3bdca68d12e8aa2cfca","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-16 06:28:41.000000000","message":"Patch Set 18:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/18/20250516140545/NEC-V-ISCSI : FAILURE in 48m 44s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/18/20250516140545/NEC-V-FC : FAILURE in 1h 22m 53s","accounts_in_message":[],"_revision_number":18},{"id":"57d2482d671a609f5d702e265a087159dc002757","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-18 03:13:55.000000000","message":"Patch Set 18:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/18/20250518104355/NEC-ISCSI : FAILURE in 57m 46s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/18/20250518104355/NEC-FC : FAILURE in 1h 29m 58s","accounts_in_message":[],"_revision_number":18},{"id":"777115ef260740103976b17f6615e4d51f783e95","author":{"_account_id":36897,"name":"Girish Chilukuri","display_name":"Girish Chilukuri","email":"Girish.Chilukuri1@partner.ibm.com","username":"girishchilukuri1"},"date":"2025-05-21 07:40:08.000000000","message":"Patch Set 18: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"5759a9560ea15bd8db8b6097207e10615965eee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-09 15:15:10.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"ce4c5a56ad6df000aea9b650db93441e3c20bb09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 15:15:19.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 912060,19","accounts_in_message":[],"_revision_number":19},{"id":"f252953a88181aa83371f6a8926588614399c409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 15:15:19.000000000","message":"Patch Set 19: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 912060,19","accounts_in_message":[],"_revision_number":19},{"id":"ce4257c3547a07c07cd3aa2d070faf8a81dfb2b1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-09 15:15:27.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev-third-party-ci/openstack/cinder for 912060,19","accounts_in_message":[],"_revision_number":19},{"id":"a844e9c988d29412396af3a0ec03bcbc12a1572f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-09 15:15:33.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 912060,19","accounts_in_message":[],"_revision_number":19},{"id":"631f47bfd6fa01315578f0f80ed41d03276f17e2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-09 15:15:59.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":19},{"id":"cdb9b3470456656b41d503952e67d91d6ba0a67e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-09 15:16:00.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":19},{"id":"7f457dffd373250c77863fc7d0067713faa49488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-09 15:20:30.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"056a8b99f282f4fa5166f7c28fde08913d327bd6","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-09 15:20:42.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"c448a8d9d3e05dab7c07bce7d65751118cbfcdcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 16:00:02.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/960210ec5e044151933a051a7a007ab5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b6d7d97c9574e76ba54fba0ebbcdecd : FAILURE in 25m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7fb3e68d543344ea8281f786a754ea77 : FAILURE in 15m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/064bf2e880e340f1835990375e0a4059 : FAILURE in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"2dd9a9ea8d702cf00bfa7991f893f1d7d38ba310","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-09 16:20:24.000000000","message":"Patch Set 20:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2d08e9eb4e9a4fb29fdab387c08ba447\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b6e4b2df40334128960ec9f0462ef765 : NODE_FAILURE Node request 200-0000065552 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b0a61a37b0e34e92806b77e007ea7f07 : NODE_FAILURE Node request 200-0000065553 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f2fddec9b6754ba29ad5a4dc0fab7de5 : SUCCESS in 50m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f2bbf969257c41d1a2f655772cbf57b6 : NODE_FAILURE Node request 200-0000065555 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ede76cacac7b4ca982fd71f2ac88fbe9 : SUCCESS in 47m 33s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a60ee2e1ac8487892d1079395a391fb : SUCCESS in 57m 55s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9487bb662f2941979d1fa06796681d2d : SUCCESS in 59m 21s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6421776f47ac4f0f84747fa72ed16aca : FAILURE in 49m 04s","accounts_in_message":[],"_revision_number":20},{"id":"9a20f0c0189b719eaa3fc4c4a49bc0256f0520d1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-09 17:10:07.000000000","message":"Patch Set 20:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-20/FC/ : UNSTABLE in 2m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-20/iSCSI/ : FAILURE in 1h 49m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":20},{"id":"35531e1d05991e772e8384ba1a687a27f7da104d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 19:06:37.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d132555400d4c9e9f003fe017784ecc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489c3a37f25f4148a16cfcd5bca99ff8 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd12bbbcde904a80b3d974ff7cd67ed8 : FAILURE in 11m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1040970ae3714c358e0a87fb5aa890b4 : FAILURE in 12m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e84e79e786934e609232dd8b7a6ffa3c : FAILURE in 14m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d59117d63b4daab775cc77dae47249 : SUCCESS in 14m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/98e6c6edfdc94e49b1ed3ee5c866c5c0 : SUCCESS in 1h 00m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d253e29dcfb947caa5f74b4fbea6ac3b : SUCCESS in 57m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/baa15cf60a8d4fb49bee7e865b61799f : SUCCESS in 1h 26m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26cac3b73afa43d58c9cf6c1313f11af : SUCCESS in 1h 00m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e37fddeffb734707b455d046e65378ea : SUCCESS in 20m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/640d4d42a68d4cb7ae750d22443bfd21 : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5d341f3010fe4bd9ab121fb1e291bd48 : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88cb9ee05b3c454e9fe83f81badb6e73 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b917f322615476aaebc4dec9a6caf1f : SUCCESS in 3m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/823d8cd4386f4e53bff19556ddf5fd7a : SUCCESS in 8m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d152b17b9e924b9884b62ccdfe95eb89 : FAILURE in 37m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b87f8f57e01044bc909870b2de6f977c : SUCCESS in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a5f1e8b62fb94800a8d45efae3d5b17f : SUCCESS in 1h 50m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b7104f84e8944e5e8cf34a8849c02c41 : FAILURE in 2h 48m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6424209020aa42f0a821d7f707725eac : SUCCESS in 1h 51m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1555773003c549769878e57d2dce355a : FAILURE in 10m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/952b061b91cb4aadb760b430cbf4415f : SUCCESS in 33m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5096b587659c44c9bd47c1c457698d44 : SUCCESS in 1h 00m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d9048d3b9ba24159993e99f6d3d62278 : SUCCESS in 1h 32m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/43f2cf9a005b4f92ad5e9f4faf9812c1 : SUCCESS in 43m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ba7ef717ad694d0583c58c4942b510b2 : TIMED_OUT in 3h 33m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2869ce254ff542b8b53eebf2d9f26bbc : FAILURE in 22m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d3472aed61e4855b1c8c0f0b2ca6b83 : SUCCESS in 1h 41m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0759f797559147878bfae81df4507f3f : FAILURE in 1h 24m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d9a754662a14db88324a00f1a871e71 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":20},{"id":"e52daed3770b1843fa345c785a23a6317729d184","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-09 19:35:34.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 46m 12s","accounts_in_message":[],"_revision_number":20},{"id":"ed62752d8b96aadbee6f109c5308250976d1a06f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-09 21:01:15.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 18m 50s","accounts_in_message":[],"_revision_number":20},{"id":"0ec38b86313c31798c305331a2a30be94b75168a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-09 23:09:04.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1a8b1bf14c674953b292b8a26ba81831 : SUCCESS in 1h 18m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bb9d4adfb16a4841831af45ef1512d2a : SUCCESS in 1h 18m 40s","accounts_in_message":[],"_revision_number":20},{"id":"90c357e71193f4aa30400bd216c4cd10f3520d18","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-10 00:06:25.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"e60d6f343d478e18940b171993e64697534852f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 00:48:58.000000000","message":"Patch Set 19:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1046/.","accounts_in_message":[],"_revision_number":19},{"id":"c6bd06bed474d534d4e477fd8f7b2976691d617a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 00:58:30.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/922/.","accounts_in_message":[],"_revision_number":19},{"id":"a9fefe6184773d9152459e50c7a40574d71f542b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 01:02:33.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1047/.","accounts_in_message":[],"_revision_number":20},{"id":"fab1077113137a19350e1d32477e18e8bb7fe7d4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 01:11:54.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/923/.","accounts_in_message":[],"_revision_number":20},{"id":"704d79be9e1a3b5b585b9a351d0419bf25b6d0d1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-10 01:32:53.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"b77b8c93252a8d87bf1a50fd62b6c2fccb8b1be2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-11 05:03:11.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/20/20250611092024/NEC-V-FC : SUCCESS in 4h 33m 44s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/20/20250611092024/NEC-V-ISCSI : SUCCESS in 4h 42m 44s","accounts_in_message":[],"_revision_number":20},{"id":"d19e53a6bd3dc4d76f95ff809c72f0cfb2a64612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-07-22 14:06:43.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"7226d070975b3f552a6cca1a15b85fc2afc98a71","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-07-22 14:08:53.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"253e818f30f0658a614b027b1ee7113070c46eb1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-22 14:09:04.000000000","message":"Patch Set 21:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/33/","accounts_in_message":[],"_revision_number":21},{"id":"dcf0f16c756b08d96acea63ab68c85dd8599caae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-22 14:18:29.000000000","message":"Patch Set 21:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/21/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/21/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":21},{"id":"7d3e12213b7c2112782f867d1294dbed83bdc072","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 14:50:18.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b205c405fa04a128d6446724a0bfbe4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97a25a89dac74f72b6d504ff601ac477 : FAILURE in 20m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a37911c13301475a8e3d58975f9c70f6 : FAILURE in 22m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/203649777e4f42459b3cd1065e2aebda : FAILURE in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"b4ad3b053d306a7ff23b85ea6c73be4dddd54ac6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-22 15:01:01.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : FAILURE in 52m 33s","accounts_in_message":[],"_revision_number":21},{"id":"b5b01c4c5c01f18d9605237f9df1812b3895250b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-22 15:07:05.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4af721a53c864d9db75308b931abe428 : FAILURE in 55m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/168e84c69a254cf2a64584f4d1e42028 : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":21},{"id":"0b08a56912558de665c1b2a570e8bafaec819a18","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-22 15:08:23.000000000","message":"Patch Set 21:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1298/.","accounts_in_message":[],"_revision_number":21},{"id":"0cee3725d3a7c1bce12ab0b9a1ee38c09ab0f00b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-22 15:12:06.000000000","message":"Patch Set 21:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1430/.","accounts_in_message":[],"_revision_number":21},{"id":"6660c25b630bacd338663802bfb4f6ac8996bfb2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-22 15:18:45.000000000","message":"Patch Set 21:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a1ead23551014218ab4e85732bbedfdb\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7aecbe4db6bf4353a19247cfa4cf6864 : NODE_FAILURE Node request 200-0000076189 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/60406ed5ae654d9c8c2d71f3e062c2da : NODE_FAILURE Node request 200-0000076190 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/be82f8fe27174dce9b9c8b234ff75b8f : SUCCESS in 1h 00m 57s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a2d9e9aa2f0142c6b1e0243a76027a0c : FAILURE in 1h 11m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1b8b4c0291a043a39431c9dea288bef0 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":21},{"id":"c650d7f732bda1b34efb1baa8b7421eec11215a8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-22 15:33:28.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 25m 16s","accounts_in_message":[],"_revision_number":21},{"id":"43904250b30478970b1aa592869ba2bc00b37b06","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-22 15:57:00.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"ca227d080aec5374e7ee55bd1e57a919978bd37a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 16:13:21.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06771f08ff2e421f89c48d3780bce2a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50dcc15f91ba4f42b8ec4f4bb0a1be75 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4392c3934c5147829e17796dec0419cc : FAILURE in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f422825648ba400dacec25e05de17358 : FAILURE in 10m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/057e2f0ca4154c37b01c5a14f0220404 : FAILURE in 13m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d795f32cb6c458395b009234064b882 : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/b683490c5e9c4b8c9642719cb3460942 : SUCCESS in 55m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bdcecdbb435a4e0fbb372f438591e7bf : SUCCESS in 34m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e8acd6f6a1c4c8aa5e66b9f229b05b3 : SUCCESS in 1h 20m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ec2b56e99cf4936a0625de9cf2df387 : SUCCESS in 50m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d87c80e1acbb4389987e5ec8009fc352 : SUCCESS in 11m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/78df48df30ae4defa1af647692549832 : FAILURE in 25m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a91eed9d8c84624a72521f408f4ea48 : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b88c1b4f473140959b6e88a614643d08 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3eb85152bcc445a59aacfadfba6b489d : SUCCESS in 5m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aa42f7a89a534f6ea200f0acdb47d09c : SUCCESS in 7m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c64b1759baa642afbafd39bae8b95501 : FAILURE in 34m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9e950d4d93748d5a72a1556bd8dab9b : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/701ea90206ef4c72937be9cd9f472979 : SUCCESS in 1h 56m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf1fd656c1274c83a7bef86f8a9b86e8 : FAILURE in 2h 05m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d308fe05d03343e2bda1fbfd0d66c54e : SUCCESS in 1h 34m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4ac2750892c44b74aa6a87803dcf1691 : FAILURE in 13m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d8f95f4616284bf48246d4c14f56ccdc : SUCCESS in 31m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/580a11d2cd4b424a9b44d9684e941eef : SUCCESS in 1h 12m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/40d6884d624d4fb1a5626908455951ee : SUCCESS in 1h 27m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ed8f1e051bcf4a6a84d42c119169fba2 : SUCCESS in 27m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/de40ca20abd241a0adbb3b34ea25aa70 : SUCCESS in 1h 25m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b0791769393b4af4ba31744a9511590a : FAILURE in 21m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d01e55f843e449380219a0265ae8d5e : SUCCESS in 49m 37s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/730fb807e23b426c85bfa21419cda728 : SUCCESS in 1h 30m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6804f01130e4a8eb9799af788bdd47d : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":21},{"id":"47eb0d242abf11a599fb7bca5eaf50d978bb0840","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-22 16:46:43.000000000","message":"Patch Set 21:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-21/FC/ : FAILURE in 2h 39m 39s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-21/iSCSI/ : FAILURE in 1h 39m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":21},{"id":"50d2a8bf107adea8872403d4396ded5bddaba667","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-25 06:30:08.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"743bebec2c5fe96cb8232a746c90c693b79bd29e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-25 06:36:31.000000000","message":"Patch Set 21:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/46/","accounts_in_message":[],"_revision_number":21},{"id":"7863c28d993f77632ed662866cf42449845b1272","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-25 06:45:55.000000000","message":"Patch Set 21:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/21/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/21/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":21},{"id":"fae51b44e113634fe750ce9201fb7db1e7d18493","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-28 13:13:49.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/21/20250728162649/NEC-V-ISCSI : SUCCESS in 5h 07m 21s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/21/20250728162649/NEC-V-FC : SUCCESS in 5h 46m 57s","accounts_in_message":[],"_revision_number":21},{"id":"9cd9b904d42ad7b65ce45bcc9411aab1d197d431","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-29 07:25:51.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/21/20250729100903/NEC-ISCSI : SUCCESS in 5h 48m 42s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/21/20250729100903/NEC-FC : SUCCESS in 6h 16m 46s","accounts_in_message":[],"_revision_number":21},{"id":"95fa352e5019dbce0ca87f7b0a54bdd49547eb3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-07 13:55:40.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"5c84f6c98fdaa72fd7411efc476383cc2ba14f9a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-07 13:55:53.000000000","message":"Patch Set 22:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/123/","accounts_in_message":[],"_revision_number":22},{"id":"e333c214c1abc64d86fc050161fe0d391b63c64b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-07 13:56:46.000000000","message":"Patch Set 22:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/22/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/22/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":22},{"id":"c8e1c77abf93630318c05b670c050112855acb44","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-07 14:45:14.000000000","message":"Patch Set 22:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7945faa38cde43b1a4bd7209d5f0e2b5\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b63d3fae1454921b4a9a5b630dd27a9 : NODE_FAILURE Node request 200-0000078267 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/75dc592cc15f45bd8f5d13dcc897a37a : NODE_FAILURE Node request 200-0000078268 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a9f68df848b147339587832aa7d562aa : SUCCESS in 49m 01s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4fc25b9da40b479a88bba96cab978ff2 : FAILURE in 48m 13s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/113a58c1aeda4f71aece0fd35aebd47c : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":22},{"id":"5a57f89697ccf4df114219043bf963bd187df8bb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-07 15:10:55.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/912060 : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":22},{"id":"3aac3ac6bae51d36c2012011c1fd95c1c6e66de3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-07 15:12:11.000000000","message":"Patch Set 22:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1390/.","accounts_in_message":[],"_revision_number":22},{"id":"c660066b74567d3a9d02cf3cb5f83fe5f4a6462d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-07 15:40:06.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"8800ac2ebf4617567eeeb81a350120ee0ade380b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-07 16:35:33.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6bcd30d1d034489c8163a54906896294 : SUCCESS in 48m 18s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d6a69363c4d2477ba9c87c3df605c7de : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":22},{"id":"e641d84c137ef7c840ace0b981b994541643ffdd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-07 16:41:10.000000000","message":"Patch Set 22:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1525/.","accounts_in_message":[],"_revision_number":22},{"id":"0551742963f5ea8fe7ace819cee2e295e1a7532f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 16:50:42.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c8da49e7d15457c8a69e8ee33806dcc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c567fc14abf2437cb4b8414186092b64 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a08b8c0096c4e36babfe8a24e8abe5c : SUCCESS in 8m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0be1ee1f280e4c3b921e4a655bf74082 : SUCCESS in 11m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc9d42c1e607460dab4200a245732a19 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/badef8d68dbc41588eb11320dbfe697c : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3bdfda8af2b94215ac0da5c8d747079f : SUCCESS in 56m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c56ddb95a8948e9b39f08dc8c1dcb92 : SUCCESS in 35m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea7ba53ecd4c478787e2b937b0a006f8 : SUCCESS in 1h 18m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/424fa8b4b26544dda320e09724378091 : SUCCESS in 58m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad568a1639c74454aaf9d490e6aa0d29 : SUCCESS in 10m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/757d5a6551be4a2aa0cf6134422435ad : SUCCESS in 21m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca53501fb1814817841f848fb55daa51 : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0606dbb14fd3499eb02033227167252c : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cacd6a789a274ef6a793f153d8c32e2c : SUCCESS in 7m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1d68793138bd47639eb74ddb7821ef4a : SUCCESS in 4m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d9a951330d0348d989b638a9a03144c3 : FAILURE in 17m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac1e6c9c5f004463b38354c015686d0e : SUCCESS in 6m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1cb39988759e48b8a3c7ea94948bc4bd : SUCCESS in 1h 17m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2d28c32f991643869fe13f2dbf87a8d2 : FAILURE in 2h 37m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3a1c40a05fd44eb386a268f5d689c3aa : SUCCESS in 1h 54m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/061ed83e192347b1bfd39cc675677582 : FAILURE in 18m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a67cd292149f40d79886b5521824e8e1 : SUCCESS in 35m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a261bdb9dbc4c9fb2ef10bb74f1d5cd : SUCCESS in 1h 13m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/23d1902ad59b4179b687ec5cbd1b29bf : SUCCESS in 2h 25m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aa81d1a41aac4b6fa92832b3af765dc7 : SUCCESS in 47m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ffe14e7c6ba143fb9b61897755592a03 : SUCCESS in 1h 30m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/953878f4a869477093a431276fd54e53 : FAILURE in 13m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/12b21223c79f4e12bf076b0c417b4d62 : SUCCESS in 1h 35m 59s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/079288c35d2d4e9ca78d70174b30b8c9 : SUCCESS in 1h 32m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83806e0f50dc46f186ddeab68db08978 : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":22},{"id":"dec16335896acedffc383b8940cf5c9e60a5fa38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-07 16:56:07.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_08_2025_16_50_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":22},{"id":"8053d7c12feba507662697193959d7087087f1d0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-07 17:27:23.000000000","message":"Patch Set 22:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-22/FC/ : FAILURE in 2h 30m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-22/iSCSI/ : FAILURE in 3h 31m 21s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":22},{"id":"641c17d75ed8f025ed1814d63d40dc39b68e3111","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-07 17:59:50.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 24m 09s","accounts_in_message":[],"_revision_number":22},{"id":"46793d98447b4d281b4764d69d36333074d75869","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 19:43:22.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db9cabe6cf7457bb676ca5762d23656\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/829b2d6af74049e4bdd207f7c6cc3b58 : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4d68220a6ac4373872fa2e0982e1f2b : SUCCESS in 27m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3e71d808b46a43fab84925a52b819841 : SUCCESS in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"b4a52cf96be83cfab43bbc5ae013e1a8adbfa083","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-08 05:21:58.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"eb7c304525a32686f7e9ab0d123499b8d61086f5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-08 05:22:06.000000000","message":"Patch Set 22:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/127/","accounts_in_message":[],"_revision_number":22},{"id":"82b5e6a64471fbfbb13e405186b3dcd6bb3dc2c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-08 05:23:00.000000000","message":"Patch Set 22:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/22/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/22/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":22},{"id":"2ee9386f01bdce211bc23bedd2e9e36a96a790f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-08 07:22:30.000000000","message":"Patch Set 22:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/128/","accounts_in_message":[],"_revision_number":22},{"id":"b6304f047810924045a9aa2283591670cdbea27d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-08 07:34:14.000000000","message":"Patch Set 22:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/22/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/22/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":22},{"id":"b5c5fb2de36ae6788f35b455dad25907cf226177","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-08-11 07:59:29.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/22/20250811111855/NEC-V-ISCSI : SUCCESS in 5h 22m 37s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/22/20250811111855/NEC-V-FC : SUCCESS in 5h 40m 31s","accounts_in_message":[],"_revision_number":22},{"id":"494cb25d5de0f37d294d63e1138ac3bd92450cfe","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-12 00:04:14.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/22/20250812024153/NEC-ISCSI : SUCCESS in 5h 55m 33s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/912060/22/20250812024153/NEC-FC : SUCCESS in 6h 22m 18s","accounts_in_message":[],"_revision_number":22},{"id":"c5a0c808133ad44e4362f5aaf34ff3739a7be30a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-18 14:43:26.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"c8867e95492c05d7cd378001b3f6893a69237c53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-20 13:53:54.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"228dd739a987b5f8f9a6fd0618a7cf41c246eddb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-20 13:54:02.000000000","message":"Patch Set 23:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/225/","accounts_in_message":[],"_revision_number":23},{"id":"cd857b993523d6c62236be24675887b690328611","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-20 13:54:41.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":23},{"id":"f934174876d370c782eb6ac2c9f3be5a777483ef","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-20 13:55:38.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"1ac1af5da9ca23f14729cd67f7e51fa681798c74","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-20 14:07:48.000000000","message":"Patch Set 23:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/23/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/912060/23/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":23},{"id":"623977fa8075ec859004f959e22f25a92900e435","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-20 15:03:48.000000000","message":"Patch Set 23:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1494/.","accounts_in_message":[],"_revision_number":23},{"id":"861c66bd0ad52c53f0b31be52f118048286d9c4b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-20 15:08:00.000000000","message":"Patch Set 23:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1629/.","accounts_in_message":[],"_revision_number":23},{"id":"29875ff3fd18a571b4dc72794649998f89c17e8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 15:32:44.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcf397dc6148413aa3e2670dad1aa4bd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86a3f7a5de6d4ed3a4543c03f7a25300 : SUCCESS in 17m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be1471c70377410595eed80e8c798e6e : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/90c52dc5016c452f9dc061546550cffa : SUCCESS in 36m 26s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"ed22f9b05a59d0c032b9685270641f128983c3e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 16:02:31.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39bb5fc5af974fe2a5485f42e23a8885\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e65244b2114b4bbbbf4028d6ebce96 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/279d91a8f9bc41d78813229ed8af06aa : SUCCESS in 9m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac7e0a82ed944d39a5821c41d341b5e7 : SUCCESS in 9m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da65680a9e9847ac8cfb0e0b1a40b3f0 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/407dafa7bf6241f38430c728bd0ab13a : SUCCESS in 19m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/390874b8484140cb9e19ac24c73700d5 : SUCCESS in 50m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29dad330783a4a84a3d6a2f33a870f50 : SUCCESS in 1h 03m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f86c4c5ffe745d79b205c8226d9abbe : SUCCESS in 1h 24m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1794cfb7e9364f02a42b1aa0c2094282 : SUCCESS in 1h 09m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07ab950d750e463fae697d1cd434cd85 : SUCCESS in 16m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/226cd829881d42c985f6d453df6dc33c : SUCCESS in 26m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/99e3e29c5ecb4077bc963c8fd7f53e72 : SUCCESS in 4m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/069d4dd13d77435692170fb5b2e361f8 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a44d114215444e9094f3f92932b1b169 : SUCCESS in 6m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/26dc4d2c12af4dc3aff5b39bf68455f6 : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2fafcf647e724c17a6b6a851383ade3c : FAILURE in 37m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc9ab2ef21974575ac777705cfa9a3e3 : SUCCESS in 7m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f50e0548eae14c5ba6134a9c1750e47b : SUCCESS in 1h 59m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8114600d03d4452c89ea35140e9d67ee : FAILURE in 1h 44m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/32a9c0a8d19f461cafcb1a2a4ac48efe : SUCCESS in 2h 05m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1f5c617c96b4473f8dce8872fae524d2 : FAILURE in 19m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/43ae84cc49d8472999a0aa0f6887451c : SUCCESS in 34m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f528d008cc0c4daebee07acd2204628e : SUCCESS in 38m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b18a72f05d824810857ba94b2940ae93 : SUCCESS in 1h 31m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b9e4e43f5ac8451e9982b6b6fa5695f0 : SUCCESS in 43m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5f5f005ac0b443cfae6b529886d2acca : SUCCESS in 55m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/761e587d9b4840d9bcc8966c5511ce08 : FAILURE in 8m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa758f68a42441d0be4a57174e79d55a : SUCCESS in 51m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90f4ad028ada456caaed916ac4ef08a6 : SUCCESS in 1h 33m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fc95b6d0b134302826e58c93c3cb44f : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":23},{"id":"97ef680c0f5761848415f1e9867e121a4ea3b4be","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-20 16:42:10.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/912060 : FAILURE in 1h 27m 03s","accounts_in_message":[],"_revision_number":23},{"id":"5e5c421414b8796b7af97d54b50031958e40aae2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-20 18:58:52.000000000","message":"Patch Set 23:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-23/FC/ : FAILURE in 5h 3m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-60-912060-23/iSCSI/ : FAILURE in 3h 43m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":23},{"id":"3e7c65f5922954fc0491e9cfc4b381ca841c19b5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-20 21:47:55.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1c104c654deb48f69017c4d51d4967bc : SUCCESS in 1h 38m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8ed5534e0ae3408687cb4dcbf9b48d15 : FAILURE in 1h 59m 50s","accounts_in_message":[],"_revision_number":23},{"id":"123f7d2daae727071652e615d65ee480add47e5b","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2025-08-29 04:36:46.000000000","message":"Patch Set 23: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"e5b0d629d93fff27879a6577e83687456871939a","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-10-30 19:44:58.000000000","message":"Patch Set 23: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"96a1553783870883760d9f238cbf050ee0e0f94d","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-07 16:15:09.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"c8f2df1b866e3c2e89b51b81ee827a6e0b0c66de","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-12-19 10:12:44.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"3c902185e010568830c516eaff1daf99f22f227a","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-03-02 09:42:23.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"3edb5471006af1cd916059ca24e96c35a764964a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-03-06 19:11:46.000000000","message":"Patch Set 23: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"0ad49304a2c6dcb77bd7da78c2117e9222234aff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-09-30 11:44:10.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"35ce3906e29fa96fc6fd25e692b6a773ac8ff491","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-09-30 11:44:45.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"a418280bf64096f088b97566a465b83079a59719","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 12:01:38.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce82635576d44eeaadfd90037b6863f4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e9fb5ead0b7489d873381b23264ed28 : FAILURE in 10m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/db197737fab84fdb816714588c782a82 : FAILURE in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"ebde7e43a1e85050bba9e9e0472b8e8b1cb56679","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-30 12:01:52.000000000","message":"Patch Set 24:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4b370222407a48ef8649024b2e0fa491/","accounts_in_message":[],"_revision_number":24},{"id":"3cea156827861dcadc99857090712a4c8633e485","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-30 12:29:06.000000000","message":"Patch Set 24:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I9e1699632e7c835df0f200d7eba03c5472193500 .","accounts_in_message":[],"_revision_number":24},{"id":"3b63bb7cfc514c2813138ff9eba86d0fc43a4cf0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-30 12:35:19.000000000","message":"Patch Set 24:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4b370222407a48ef8649024b2e0fa491\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/29c014edca424852aef827f5789d41c2 : SUCCESS in 38m 35s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/97ad0b053122497a9a3f9fd464d64541 : SUCCESS in 32m 06s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfc0ace44c4c482a84273d03711de58a : SUCCESS in 50m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b6d957a525e4a6b885e4853316950d6 : SUCCESS in 33m 15s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2be6bb612e344696a8213c2d5368b2b6 : FAILURE in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"9553e609c62f2eef3c337f5e09a9d651fca497e5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-30 12:39:21.000000000","message":"Patch Set 24:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 16s\n- FC       : SUCCESS in 0h 50m 7s\n- NVMe/TCP : SUCCESS in 0h 47m 57s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/912060","accounts_in_message":[],"_revision_number":24},{"id":"3967cc735bd3821f6556662c1d57af53b3d8c5b8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-30 13:06:52.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/881f4634668c4e558819f56d08e863ba : SUCCESS in 47m 11s\n- cinder-storpool-tempest-iscsi-multipath build/46ed1bbe36ee471f87595b52b151a538 : SUCCESS in 50m 25s","accounts_in_message":[],"_revision_number":24},{"id":"1a3364807cde63e74ecca54d055157ffe18cab64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 14:20:12.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99613e641f75439bad6be146fb6dec85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b65ece019b2442e9b3ab940daf5edbc : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1db637a39a124e9db7802630cbb813ef : FAILURE in 10m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06e081fadf674796971489fee83439cc : FAILURE in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97efc67d74f48e183ec7ed6cb972480 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/4c3fcfcf11734d08a76a7ddac2a1546f : SUCCESS in 57m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7827eb607a64301987e5f7fa4ddf107 : SUCCESS in 1h 11m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a6bc4fd02574391a34f6cb830234940 : SUCCESS in 1h 22m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d6f6c8dcffb4f258da3a5f4de21d363 : SUCCESS in 31m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5e2904edd124752a73c412316bdfb51 : SUCCESS in 11m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e723e335ba074e198e253bae075b7586 : FAILURE in 24m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9d3c1b55506b482dafa1845de3a91507 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bfb2ba21db774c1dadb66b52a7234872 : SUCCESS in 6m 13s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/2ec06e2c44c647bf8f3a52269355735f : FAILURE in 11m 16s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d9d4641a084049938415e128c0fa0195 : FAILURE in 7m 19s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/43e12784eb364400af186cf66b186850 : SUCCESS in 4m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/76842eb4bd4541c0bd18ec43a1bbee6d : SUCCESS in 5m 04s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/405eaeea2c174eff9dc02ef9c53a9723 : FAILURE in 5m 31s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7396f12ffd864bccbea9ffd3d432e81c : SUCCESS in 8m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2bddfd200af349a4b859b97db0196eef : SUCCESS in 37m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aadedff7496e417086f7d41097074100 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/499b16d7f5514cee914353fa4a824639 : SUCCESS in 44m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0243780ca5ac4cdd884d0ad4db0f0f70 : FAILURE in 1h 51m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/05eeceaf64df4cc7ad1fe00e71451a1e : SUCCESS in 1h 45m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/5066e55fc843401893c8785c2cfe94c6 : FAILURE in 2h 32m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0f0e2761c49e47c2a1ab3be060480f7e : SUCCESS in 1h 56m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bf12403d4f8f4755b04df79dac761a2f : SUCCESS in 35m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cce280d22d1c49609743e5d67161ea57 : SUCCESS in 49m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48e2428071e2442ca1429ae4ee2888dd : SUCCESS in 1h 37m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b40bc2aed4f7496a832125a667b4342a : SUCCESS in 29m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58e58c66ad544aad886766c09b39d24a : SUCCESS in 1h 36m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/551a02dfc9e446dab594d7f6efe31d42 : SUCCESS in 29m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":24},{"id":"0795675999ab7a9b95b127cf50cdf875a297d082","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-30 14:32:09.000000000","message":"Patch Set 24:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5595/.","accounts_in_message":[],"_revision_number":24},{"id":"c94625f1788b5e9d3ef8d6abdc99aa56ef132c9f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-30 14:58:08.000000000","message":"Patch Set 24:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5772/.","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"d6dca26a11383a2ed7b93e90aae04afcec798c2f","revisions":{"3ce9f834db54dc0d158844a825aa9d006a762550":{"kind":"REWORK","_number":1,"created":"2024-03-14 11:28:30.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/1"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 11:27:57.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ce9f834db54dc0d158844a825aa9d006a762550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ce9f834db54dc0d158844a825aa9d006a762550"}]},"branch":"refs/heads/master"},"2ce30086b814b35cb93f0cc2577ca3303d8975cf":{"kind":"REWORK","_number":2,"created":"2024-03-15 10:24:33.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/2"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-15 10:24:33.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ce30086b814b35cb93f0cc2577ca3303d8975cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ce30086b814b35cb93f0cc2577ca3303d8975cf"}]},"branch":"refs/heads/master"},"3aaef3c527c9a4c45effb23dfb47ab1a759619ac":{"kind":"REWORK","_number":3,"created":"2024-03-15 10:31:08.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/3"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-15 10:31:08.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3aaef3c527c9a4c45effb23dfb47ab1a759619ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3aaef3c527c9a4c45effb23dfb47ab1a759619ac"}]},"branch":"refs/heads/master"},"033c52a5c402e8ef04fb0f6b9500758fcec8810c":{"kind":"REWORK","_number":4,"created":"2024-03-15 11:03:27.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/4"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-15 11:03:27.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/033c52a5c402e8ef04fb0f6b9500758fcec8810c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/033c52a5c402e8ef04fb0f6b9500758fcec8810c"}]},"branch":"refs/heads/master"},"58cb5392db54a3b43bde8d04e6e4d8d1daa75edf":{"kind":"REWORK","_number":5,"created":"2024-03-18 05:39:25.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/5"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-18 05:39:25.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58cb5392db54a3b43bde8d04e6e4d8d1daa75edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58cb5392db54a3b43bde8d04e6e4d8d1daa75edf"}]},"branch":"refs/heads/master"},"f0593b91d249843d4169d38b76cbe8d269c5aaef":{"kind":"REWORK","_number":6,"created":"2024-03-18 07:54:09.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/6"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-18 07:54:08.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0593b91d249843d4169d38b76cbe8d269c5aaef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0593b91d249843d4169d38b76cbe8d269c5aaef"}]},"branch":"refs/heads/master"},"e6f35de0dfc144448cd2972e955a3d3792c9f2d1":{"kind":"REWORK","_number":7,"created":"2024-03-18 09:01:03.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/7"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-18 09:01:03.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6f35de0dfc144448cd2972e955a3d3792c9f2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6f35de0dfc144448cd2972e955a3d3792c9f2d1"}]},"branch":"refs/heads/master"},"cb9d8a1ab7d8dbacbfb60ff71fdf335bf2f0d369":{"kind":"REWORK","_number":8,"created":"2024-03-27 12:58:09.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/8"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-27 12:58:08.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb9d8a1ab7d8dbacbfb60ff71fdf335bf2f0d369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb9d8a1ab7d8dbacbfb60ff71fdf335bf2f0d369"}]},"branch":"refs/heads/master"},"d411c59dbca6b3f5a5e13bdc689beb6bac826ed7":{"kind":"REWORK","_number":9,"created":"2024-03-29 06:31:14.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/9"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-29 06:31:13.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d411c59dbca6b3f5a5e13bdc689beb6bac826ed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d411c59dbca6b3f5a5e13bdc689beb6bac826ed7"}]},"branch":"refs/heads/master"},"11759e34e1dc41a4a04848f6214efe42fbe5a8f0":{"kind":"REWORK","_number":10,"created":"2024-03-29 06:45:59.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/10"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-29 06:45:59.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11759e34e1dc41a4a04848f6214efe42fbe5a8f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11759e34e1dc41a4a04848f6214efe42fbe5a8f0"}]},"branch":"refs/heads/master"},"31d874fe9321e9b94327250a8af2adfd28fd5b35":{"kind":"REWORK","_number":11,"created":"2024-03-29 09:22:39.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/11"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-29 09:22:39.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31d874fe9321e9b94327250a8af2adfd28fd5b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31d874fe9321e9b94327250a8af2adfd28fd5b35"}]},"branch":"refs/heads/master"},"2652ee4872993fd0da7d02149320ccdb28a848be":{"kind":"REWORK","_number":12,"created":"2024-03-29 09:25:48.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/12"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-29 09:25:48.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2652ee4872993fd0da7d02149320ccdb28a848be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2652ee4872993fd0da7d02149320ccdb28a848be"}]},"branch":"refs/heads/master"},"1376b935d74c5d85eda981f38894a7b11bb2c844":{"kind":"REWORK","_number":13,"created":"2024-03-29 11:14:41.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/13"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-03-29 11:14:40.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1376b935d74c5d85eda981f38894a7b11bb2c844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1376b935d74c5d85eda981f38894a7b11bb2c844"}]},"branch":"refs/heads/master"},"04438849879497a1219b33ac53b77d9f69b7bc1b":{"kind":"REWORK","_number":14,"created":"2024-04-01 06:58:00.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/14"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-04-01 06:57:59.000000000","tz":0},"subject":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[SVf]: Enable support to retype ‘iogrp’ of an ‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04438849879497a1219b33ac53b77d9f69b7bc1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04438849879497a1219b33ac53b77d9f69b7bc1b"}]},"branch":"refs/heads/master"},"629f8e9563b25996559cd869083830e21985552a":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2024-09-18 14:03:23.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/15"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-09-18 14:03:23.000000000","tz":0},"subject":"[IBM SVf Driver]: Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver]: Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/629f8e9563b25996559cd869083830e21985552a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/629f8e9563b25996559cd869083830e21985552a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ae1035acb4d90b7d6e0af0079bb793d9dbdbc218":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2024-09-18 14:03:52.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/16"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-09-18 14:03:52.000000000","tz":0},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae1035acb4d90b7d6e0af0079bb793d9dbdbc218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae1035acb4d90b7d6e0af0079bb793d9dbdbc218"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5822fb5b783528d2ae021467e080acfc64552b7a":{"kind":"REWORK","_number":17,"created":"2024-12-26 08:49:26.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/17"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2024-12-26 08:49:26.000000000","tz":0},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5822fb5b783528d2ae021467e080acfc64552b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5822fb5b783528d2ae021467e080acfc64552b7a"}]},"branch":"refs/heads/master"},"60a12796f65a759805212234077f0f443a07c603":{"kind":"REWORK","_number":18,"created":"2025-05-14 15:11:07.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/18"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-05-14 15:11:07.000000000","tz":0},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60a12796f65a759805212234077f0f443a07c603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60a12796f65a759805212234077f0f443a07c603"}]},"branch":"refs/heads/master"},"533e19de7104ebf2f739833797bb4a799315c5ea":{"kind":"REWORK","_number":19,"created":"2025-06-09 15:15:10.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/19"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 15:15:10.000000000","tz":0},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/533e19de7104ebf2f739833797bb4a799315c5ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/533e19de7104ebf2f739833797bb4a799315c5ea"}]},"branch":"refs/heads/master"},"0fab84bb7e3fd8c0663e841b95c29ea942e33204":{"kind":"REWORK","_number":20,"created":"2025-06-09 15:20:30.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/20"}}},"commit":{"parents":[{"commit":"deef0d4d28d6a1ba84b9f3af49efdd85042f68ff","subject":"Merge \"[SVf] : Enable support for mirror-pool option for replication volume-type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/deef0d4d28d6a1ba84b9f3af49efdd85042f68ff"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 15:19:45.000000000","tz":330},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fab84bb7e3fd8c0663e841b95c29ea942e33204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fab84bb7e3fd8c0663e841b95c29ea942e33204"}]},"branch":"refs/heads/master"},"9ade36e9e039d986deb7d4db88bbef91b48924c1":{"kind":"REWORK","_number":21,"created":"2025-07-22 14:06:43.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/60/912060/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/21"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-07-22 14:06:31.000000000","tz":330},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\nSigned-off-by: nimeshdesai \u003cNimesh.Desai@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ade36e9e039d986deb7d4db88bbef91b48924c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ade36e9e039d986deb7d4db88bbef91b48924c1"}]},"branch":"refs/heads/master"},"7de012f3d2ce6f2084413a4c9f1840bd6e9f6d4f":{"kind":"REWORK","_number":22,"created":"2025-08-07 13:55:40.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/60/912060/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/22"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-08-07 13:55:26.000000000","tz":330},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\nSigned-off-by: nimeshdesai \u003cNimesh.Desai@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7de012f3d2ce6f2084413a4c9f1840bd6e9f6d4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7de012f3d2ce6f2084413a4c9f1840bd6e9f6d4f"}]},"branch":"refs/heads/master"},"ed48542f1154682f5a738d3703443af10a3af64d":{"kind":"REWORK","_number":23,"created":"2025-08-20 13:53:54.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/60/912060/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/23"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-08-20 13:53:33.000000000","tz":330},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\nSigned-off-by: nimeshdesai \u003cNimesh.Desai@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed48542f1154682f5a738d3703443af10a3af64d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed48542f1154682f5a738d3703443af10a3af64d"}]},"branch":"refs/heads/master"},"d6dca26a11383a2ed7b93e90aae04afcec798c2f":{"kind":"REWORK","_number":24,"created":"2026-09-30 11:44:10.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/60/912060/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/912060/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/912060/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/912060/24"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-03-14 10:42:02.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2026-09-30 11:44:10.000000000","tz":0},"subject":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an ‘in-use’ volume","message":"[IBM SVf Driver] Enabled support to retype ‘iogrp’ of an\n‘in-use’ volume\n\nRetyping ‘iogrp’ of a volume makes a ‘rmvdiskaccess’ call\nto remove the volume’s access to the old_iogrp.\nIncase of bootable ‘in-use’ volumes, ‘rmvdiskaccess’ is being called\nbefore the attached_host could discover the new path for the volume.\nThis results in crashing of the host as it loses the access to its\nbootable volume.\n\nCloses-Bug: #2055149\nChange-Id: I9e1699632e7c835df0f200d7eba03c5472193500\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6dca26a11383a2ed7b93e90aae04afcec798c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6dca26a11383a2ed7b93e90aae04afcec798c2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
