)]}'
{"id":"openstack%2Fcinder~913204","triplet_id":"openstack%2Fcinder~master~I8fdbc8087115761d796f6f5fd85253df47a35049","project":"openstack/cinder","branch":"master","topic":"fujitsu-driver-update","attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-08-06 06:48:17.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2025-03-26 08:45:20.000000000","reason":"Someone else replied on the change"},"37834":{"account":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"last_update":"2025-04-04 05:41:38.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8fdbc8087115761d796f6f5fd85253df47a35049","subject":"Fujitsu Driver: Add support for retype volume","status":"NEW","created":"2024-03-15 06:09:06.000000000","updated":"2025-04-09 14:14:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f5357dd5a78441052407caeae29f409de3537842","_number":913204,"virtual_id_number":913204,"owner":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-04 02:33:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 06:09:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 06:33:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 06:35:28.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-15 07:08:19.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-15 08:42:43.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-15 08:53:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-15 10:26:22.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 20:07:20.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-16 01:34:47.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-18 03:19:07.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 06:51:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 07:45:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 08:56:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-19 15:27:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 22:35:11.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-20 08:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-20 11:12:04.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-04-15 15:12:09.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-05-27 14:13:04.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-27 17:19:42.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-28 15:17:45.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-05-28 15:52:22.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-05-29 02:05:33.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-06-14 10:00:37.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-06-14 15:34:53.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-06 06:48:28.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-02 04:48:42.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-15 04:53:39.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-03-21 22:02:28.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-24 05:25:19.000000000","updated_by":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"reviewer":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"state":"CC"}],"messages":[{"id":"1bb1c5e5c0fe85d3a7f6e995d636ac1aef66c34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-15 06:09:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"031c874b7d685d7014f5b3b5e2a690f76416e27e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 06:09:14.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16639/","accounts_in_message":[],"_revision_number":1},{"id":"66f6fae43c704955f2e1d54ae8585662ceec321a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 06:09:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16639/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6941b24e985c4c1c9c51e9e634273f8b293aae16","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-15 06:33:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 19m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2d51c47173f50ebb07ee570110012b2c57eecd7e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 06:35:28.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/67c926d6982f4688b176d53a5e09893c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/743d9e84f95b42da9423fa42d25a24e9 : NODE_FAILURE Node request 200-0000002784 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"7fcd92cfad07155ae62ae9d8ae2e8b8ba0af01b0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-15 07:08:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 51m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ba08d9fe7ee5ceb6b919d4656785bac0de7705dd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-15 08:05:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 51m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bae598911e54e7625867884308d0bcda6bf474cd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-15 08:42:43.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-1/FC/ : SUCCESS in 2h 33m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-1/iSCSI/ : SUCCESS in 2h 2m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"6f83db51ed3be6229ae6d35520f023fc5f90bcb9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-15 08:53:34.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/1/check/DellEMC_Unity_FC/cefc74e/DellEMC_Unity_FC/1508 : SUCCESS in 1h 16m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/1/check/DellEMC_Unity_iSCSI/d33e371/DellEMC_Unity_iSCSI/1540 : SUCCESS in 1h 34m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d0bac762828ac640876e5e85a29f595e4e37cd0","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-15 10:26:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/1/check/fujitsu-eternus-dx-iscsi/eca50fa : FAILURE in 3h 50m 42s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/1/check/fujitsu-eternus-dx-fc/bb6fb19 : FAILURE in 2h 49m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b7b754c7dee11a8bfae6ef698ff22748a930f283","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-15 20:07:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/1/20240316014437/NEC-V-ISCSI : SUCCESS in 3h 19m 38s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/1/20240316014437/NEC-V-FC : SUCCESS in 3h 22m 40s","accounts_in_message":[],"_revision_number":1},{"id":"66d959985f4b5c3b159839bf92248e4a61b9fd50","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-16 01:34:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/04/913204/1/ : RETRY_LIMIT in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a6c87c5035ed1aadee8f22e100bc292236bb00da","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-18 03:19:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/1/20240318034152/NEC-ISCSI : SUCCESS in 4h 11m 53s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/1/20240318034152/NEC-FC : SUCCESS in 8h 37m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f2daf85796f02fdbd0b5d14e64ff208ffcdc6d31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-19 06:27:27.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fedb5ba7dcddf9c04142d1b42cc0c1727f367b44","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-19 06:27:39.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16674/","accounts_in_message":[],"_revision_number":2},{"id":"6b2e0b2b028b12f8a8b2f1b1e3060057d338e98f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-19 06:27:41.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16674/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a5f8559d653eef596cadd8d1ea29352149fbcd5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 06:51:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c821e03215bf400b9074a8de410c2ede\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1963492989949b2a47979347be52c7b : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0a254b5ce4084f27b801d73d230ccb1f : FAILURE in 14m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1288bb5024ad4c12802ca75f2fc71bd8 : FAILURE in 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"921f39af7cfe93716ebb2bd5f200b8ba8684c724","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-19 07:45:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c05ecfe700bd40e1a9b466c35ec7f1b8 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3886cc8edfb44d1091535e61b90246f5 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"42b194f974bdfe37e7a764db1d19fd2a8090f556","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-19 07:49:21.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/0265d8d0d1294b4194eebabc0a9727c1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2adfe16f6c3d493986ba417826a644d3 : NODE_FAILURE Node request 200-0000002829 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"5885b78557779e176884dc56e3e64f79ad6426da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 08:56:48.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4a2ac57d5104164922c39cfe170a358\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08784d8e061240618d0354eb63298ecb : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41943d989d0e4455ba8e9319c7f2cdbd : FAILURE in 32m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b52e921f3134e0d912a91bd1c686ce6 : TIMED_OUT in 46m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2386c9fe5bf4ed3852002dde42e295c : FAILURE in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/d776a96ec7bc4355b5c08625b0cab259 : SUCCESS in 1h 03m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/50313dfe982e4815a933fa986a64d676 : SUCCESS in 58m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4005da678c914489a44514a7a17efc51 : SUCCESS in 1h 25m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66c53f27adb74b55bc21e8055497b785 : SUCCESS in 1h 04m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5239a5c1dc614df99fd554c79ed7e97f : SUCCESS in 16m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1c1e022c95a940ddb2ec1e1a18fe2bd1 : FAILURE in 30m 03s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5af8f745d42849b4a6e46d3036456ad5 : FAILURE in 39m 38s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/79d9d2a6d0604430b982b275910105b5 : SUCCESS in 6m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e6ea411178784c778303c02d09452585 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e554df1e9b62412f8343e704cd36647e : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fdd80f7c2f204bb68a2627656ed36dd8 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2dec58fc6a074a3883276d8dfd17e037 : SUCCESS in 45m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/813219822b004fb88209ab773cdfcf4e : SUCCESS in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5625c433f75c4dc78d755cc88a881bed : SUCCESS in 2h 17m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ade58a9fed084848a277f62b6339deb3 : FAILURE in 2h 04m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/726f9a3ec6874fc9b8c5340466a92aab : SUCCESS in 1h 46m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/001d5a49b490497a8462e3ce516e2dc1 : SUCCESS in 2h 06m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eea252a72e094636b65530e776956ffc : SUCCESS in 39m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/539b725d7ee54fc387797a59f2e43eb9 : SUCCESS in 1h 20m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/936169d1b1ff4767b5b207ca6dc79c37 : SUCCESS in 1h 40m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/361f068a686d4e4bb79b06401bb691cd : SUCCESS in 1h 00m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d41fdf2b031b4ea49dc96866f99a8416 : SUCCESS in 1h 16m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/60e15203997f4185aa4f200042041c9a : SUCCESS in 1h 40m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9dbb7a80d9f749a4b697df7bbc1302e2 : SUCCESS in 1h 36m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64fd1d5ba4234674b277f6fbf347906d : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8444d0ea4869673ee531917f14ff9b60cb39b039","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-19 09:42:50.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"59db4a0f5680e8d53ad0525edf7afa3be00cca40","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-19 10:03:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f457c216d0444a6db88fc3fdf3d77f04 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/875b437c7d8645028a4eacaac824c6f8 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"1d078aaf00913626717914a4068fc33093fa806c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 10:07:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b980cf2c9cb040f1be212c559e8a3e50\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/baa93e66560544429fa4a17268e41347 : FAILURE in 12m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/16a3ede2c196426c9efeef80e03d4cb5 : FAILURE in 22m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d2452811d4b47468b141bcf244a74f3 : FAILURE in 12m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c622482465b07484cae7c85b5c3925613fcfa2a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-19 10:19:02.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c2151847919f4da6bd3dac8e0aa83e08\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/518511a53853471ea6f1c8cdc3d15fab : NODE_FAILURE Node request 200-0000002839 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"0ed503bcff297cb128f4c5ca855bbaf9c2a4b489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 12:44:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6cc0196ad714f4a85d350605f247719\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7259a1ff169f4543aa880dd85d55bd4e : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/835a2ed18ce4482ab32dbdf4b2654f88 : FAILURE in 20m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c6db50112bf432d93684b35c2f71d9d : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2cbde261604b43b3d154939693f93f : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e635fc6fc8bd4c4eb8ec98b7836d05d7 : SUCCESS in 58m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/576e2f1970a3404099ab21e674c4fb4a : SUCCESS in 58m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b12eff468984ce38873d67db9f10487 : SUCCESS in 1h 22m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a99d9bd741504385862121a70ec61758 : SUCCESS in 58m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf4ae84a92344262a0349aec6982a81e : SUCCESS in 13m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/386d5cd1dbef4f15bcdffc819e7155eb : TIMED_OUT in 42m 40s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f318d24a7d4a44caaadacf0f7c95a34e : FAILURE in 17m 01s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/616341c062214f1e8d209c192b2c9acd : SUCCESS in 6m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ffa7af10f826479bbc9214288257c758 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bd6720b9b28644d484fa6a4956c05f10 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7727c305ea2f42f59983f5c95e7f3904 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a2b590dfc14a493db07f258c49981464 : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e4c6042419c44a33b70b4ff3f68e63c6 : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f5f93cecd3f4c6f9cadbdac76871e15 : SUCCESS in 1h 46m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ff6d3f72dc4e41bcbee531626765bd12 : FAILURE in 2h 53m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/72d9cc7b7057439cae719f23dfa65ca0 : SUCCESS in 1h 40m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/33bb23357a824aac948e2b4b1b0a02b7 : SUCCESS in 2h 09m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/56f4909ab2d6463880462a644d62c962 : SUCCESS in 44m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c95ff6ecd30b4638b632d124b71831ae : SUCCESS in 1h 06m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/385eac5fa4124cdca86d65a5a48ee581 : SUCCESS in 2h 02m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a5e68a92276947a38eff0902dade89e9 : SUCCESS in 47m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2f3b9a8d226045e3ba78fcaa18e7219b : SUCCESS in 1h 15m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/80dd8802f68d4497a6df813f6aaa8b07 : SUCCESS in 1h 42m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ac57fdc629a470aa427174e9bf7bedf : SUCCESS in 1h 29m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1b2086eafd14bd58369877e1d000ef8 : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":3},{"id":"8acb5e540cced35231ffcf222e3233acd431c312","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-19 13:42:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/3/check/fujitsu-eternus-dx-iscsi/8b71463 : FAILURE in 3h 28m 26s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/3/check/fujitsu-eternus-dx-fc/d7ca93e : FAILURE in 2h 51m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f8075ea8d2c5607541933d5b107c39f6eb7e2480","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-19 15:27:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/04/913204/3/check/DellEMC_PowerFlex_3.6/8789ac9/DellEMC_PowerFlex_3.6/1663 : SUCCESS in 1h 50m 08s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/04/913204/3/check/DellEMC_PowerFlex_4.5/2620989/DellEMC_PowerFlex_4.5/604 : SUCCESS in 2h 16m 31s\n- DellEMC_PowerFlex_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/3/check/DellEMC_PowerFlex_NVMeTCP/2571d84/DellEMC_PowerFlex_NVMeTCP/16 : FAILURE in 2h 52m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"7a5410a694340e3644e0c462a0771d052b34560e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-19 17:23:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 17m 02s","accounts_in_message":[],"_revision_number":3},{"id":"4df02d74f4e38543d91e496085fb3a1b287019cf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-19 22:35:11.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/94/.","accounts_in_message":[],"_revision_number":3},{"id":"f9b07fb275fa0296241dbaf1eecbb70d89fb2cf3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-20 06:02:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 48m 20s","accounts_in_message":[],"_revision_number":3},{"id":"0a3819eb215123a30ba543fa558887a16ffa6cf5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-20 06:50:36.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/3/check/DellEMC_Unity_FC/5248e4a/DellEMC_Unity_FC/1547 : SUCCESS in 1h 13m 40s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/3/check/DellEMC_Unity_iSCSI/04a5813/DellEMC_Unity_iSCSI/1579 : SUCCESS in 1h 41m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4fc8a06cd01d50ce480715e39f698e3130e191b2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-20 07:06:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 57m 45s","accounts_in_message":[],"_revision_number":3},{"id":"1bbb1913987927a1c63c7353d39fec45c7da6438","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-20 08:00:06.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/3/fc : SUCCESS in 0h 37m 34s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/3/iscsi : SUCCESS in 0h 48m 22s","accounts_in_message":[],"_revision_number":3},{"id":"dbd1f9967720d7001f860cd4edb49d4c93f4dc70","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-03-20 11:12:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ec6a984a9448feef82e59d57c98653d7e862c5f3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-23 17:00:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/3/20240323222944/NEC-V-ISCSI : SUCCESS in 3h 07m 39s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/3/20240323222944/NEC-V-FC : SUCCESS in 3h 30m 51s","accounts_in_message":[],"_revision_number":3},{"id":"bcabd0a510f60d72dd1c4793bdafc20e091927a2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-27 15:16:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/3/20240327193540/NEC-ISCSI : SUCCESS in 4h 15m 01s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/3/20240327193540/NEC-FC : SUCCESS in 4h 40m 16s","accounts_in_message":[],"_revision_number":3},{"id":"d12c51f41a24b4a34eb05a52cb61a46f6c837241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-04-09 10:14:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9431684a600cb20b757c1517bff35c010f9409c4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-09 10:25:37.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6bba278037034b03a92e791d930dbefe\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8e79fe18c1484e69bff28cb66f68b8ca : NODE_FAILURE Node request 200-0000002952 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"a58e64a4308bd949c5db300460a695097c796e29","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-09 10:55:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 37m 07s","accounts_in_message":[],"_revision_number":4},{"id":"e1fb3eec37ee8acd879efae892123994dd43471f","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-04-09 12:29:03.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a74143ca50bfd8a1d7bcdf7b0f5cbbc41503cee9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-09 12:44:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/188/.","accounts_in_message":[],"_revision_number":4},{"id":"9c22b352ad5785c78789f1b2ab558f328d59ceee","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-09 13:00:06.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/4/fc : SUCCESS in 0h 45m 30s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/4/iscsi : SUCCESS in 0h 51m 28s","accounts_in_message":[],"_revision_number":4},{"id":"0edbe317de19130bcb5560142fa591ef2814bd0a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-09 13:19:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/4/check/fujitsu-eternus-dx-iscsi/ab6fa6a : SUCCESS in 2h 58m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/4/check/fujitsu-eternus-dx-fc/e8c4253 : FAILURE in 2h 55m 15s","accounts_in_message":[],"_revision_number":4},{"id":"91da16b05dd4ed2a5489d6ad6507d479a132e879","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-09 17:31:14.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-4/FC/ : FAILURE in 7h 16m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-4/iSCSI/ : SUCCESS in 2h 3m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"04f7a764a63dcb063b0d5f9b749c5c615fc00de3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-10 12:07:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/4/20240410093514/NEC-FC : SUCCESS in 9h 19m 03s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/4/20240410093514/NEC-ISCSI : SUCCESS in 11h 31m 43s","accounts_in_message":[],"_revision_number":4},{"id":"cde162e318e69828753105eeb6f0fd770130de9b","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-04-15 15:12:09.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"357d230a582a1511afc25695463e244f5c767ca3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 06:33:32.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"0c9a4116ee538a5020ef0fb8091227a6de7d54b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-21 05:31:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"407cae6f50f5a5c0f6488705361aec73f00df31f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 05:46:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f1be272ca4846fdbd0b0940780d2a12\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/614226d9bed24c68a43cde1c02be4bbe : FAILURE in 13m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d3ac913cb30d45a2b399517adbcf8873 : FAILURE in 12m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c70676f82c6a9ee63054e04c99c9132caff20fd5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-21 06:46:09.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 24m 18s","accounts_in_message":[],"_revision_number":5},{"id":"c091fc38e506c7724589f7aa787197de907ca1d9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 07:46:43.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/80d394ba112746f9bf75c6158de17096\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/316fb5aaa938400fb111de06ebd38715 : POST_FAILURE in 27m 19s","accounts_in_message":[],"_revision_number":5},{"id":"40494d4e38dc88f2497b16045760541376b6e7f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 07:52:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2c6d8e107294a26aaeb971ab626558d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ea9a67ce4148108d95058e43e55178 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30ea488660c44c0b91f80ee04e2ce248 : FAILURE in 13m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cf360e5832041d1a12ca3f2e2ac4125 : SUCCESS in 9m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb897ccca3544550a8e63aa7cb097edc : FAILURE in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1525bcb0caa4dabb9e1e6052d907de4 : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/9defe1a4490849909d2628b92d30bb93 : SUCCESS in 1h 06m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce530f2fac154808ba368262c178f1ba : SUCCESS in 1h 16m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f46f75f79aff41648313e7a492a14ffa : SUCCESS in 1h 03m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e929917c059347d890a06fe654042213 : SUCCESS in 10m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c84b9d466264a5ab43a7e1906851006 : FAILURE in 36m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11b278120fec4d99bc0c4c05462320a7 : SUCCESS in 6m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6af8e84d7bee40c2b20b187a3d613504 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e76f4f1a6e564c57bc55c156b8fc35c6 : SUCCESS in 7m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c939bb9384cf42aaa532a98da60b9cef : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/58f4ac8798ea4de49396da7dcf59066d : FAILURE in 28m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51291a1718ef40bc8ee6044bbf54b8c2 : FAILURE in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/321c75e52da24c139f4ee615a52c65e6 : SUCCESS in 2h 02m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b831ec3cc24c4537b9d36f104735a255 : SUCCESS in 2h 17m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/268b4e5beb6b4f1ab84959ef120ee3cd : SUCCESS in 1h 58m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6b0dfb3ff11c43babf4bf5e25a6c1fb3 : SUCCESS in 1h 38m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1efdbdcf4f0f44b296d006bbea5687d0 : SUCCESS in 39m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/59dfe17ef21c43e1bae12f755082223e : SUCCESS in 1h 01m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/185397920cbb47659a4acea732a98c26 : SUCCESS in 1h 06m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b3e20c70cafb4fbe9742b1fd170aa52b : SUCCESS in 34m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/95ee5842616b40289477e6bd265c5a10 : SUCCESS in 1h 05m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3c99d564dbd54ee599a840df3a43668a : SUCCESS in 1h 43m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a0045b880384614aba36d915b9ece2d : FAILURE in 1h 27m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cc52c51528a4fea9dffba9911c0deb9 : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":5},{"id":"58370bb66f7204f148b6863d7cb1b971c292ac51","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-21 08:32:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 35m 59s","accounts_in_message":[],"_revision_number":5},{"id":"fc36dfefa645297f153492831a4d8109f5991fd1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-22 05:35:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/452/.","accounts_in_message":[],"_revision_number":5},{"id":"c5ac56db5e2a552e6742cf7012282f8e31d3450d","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-22 05:44:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b07bb2f64c2c83bd2a0f7af7ca1596873ef900e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 05:57:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce2c14f792a8407b872edf79493247a5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa59deaf964c495e95b3b4f992573925 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b6096c2656a48ca8e8785684e714c99 : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"899879a1a0cbe091f6da344ffb8fa4ba55157f16","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-22 07:42:32.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5fb7b6e57c9442c2ab2a31cdb8e76ae1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/509e3e322c5f46168d777db1efe5f483 : POST_FAILURE in 25m 51s","accounts_in_message":[],"_revision_number":5},{"id":"03b356ba16c719e2653b914ef91e0bb3b0c74c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 08:25:47.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77f09088095c40e3885c24da8a06582b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6af769ce9834b8dbdc4365283057bba : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5e252441b69464395ad04a93607c049 : FAILURE in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1bf09c859644f4883680349c16de0f4 : FAILURE in 12m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/abb88bb633b14cbfa8fdba5ebb95fccb : FAILURE in 13m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd333f4dc99042f2b631be9b55f9d5e2 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/194d6b1ba84442e9b37373f7e3136b6d : SUCCESS in 54m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59d2d71a530448eaa38ea07f559e3789 : FAILURE in 1h 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb9893a956a145e3b864b1850611caca : SUCCESS in 44m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d880b1a4578c4e189a13ed7255bf6496 : SUCCESS in 12m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a89370875cb94c26a25a05b5a5495931 : SUCCESS in 17m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/adf87f00738b49bab437ec413b042605 : SUCCESS in 6m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/71c0d469634d40a2a7b916bb2f1fc248 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b81bcb4ea31246ad85b6f2bb45124b95 : SUCCESS in 6m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/90527f1f9b904b99af0971b85e16047b : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9887df9936334417a25b68647bc6b032 : FAILURE in 51m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/731bec806c3c4126ac33b304ad73ff05 : FAILURE in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ec8ce00178924f11a064d4448e6a81bf : SUCCESS in 2h 12m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/141a758065ee4ad6b9506315f25d287e : FAILURE in 2h 26m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f944bf6b35c14dedba20f784431d0bca : FAILURE in 1h 59m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cf542d98553741218884f39048ac99fa : SUCCESS in 1h 41m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/793b75a355e840acbc883f970d86025b : SUCCESS in 43m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/13bc0661d4154933a6843835207732d2 : SUCCESS in 1h 12m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8f39046358704f3694dde3593587385c : SUCCESS in 1h 32m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76990d96da524540ab557961269314b3 : SUCCESS in 32m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/36a4d5f3c5f7416597bb8d3a8d83b658 : SUCCESS in 1h 03m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/74acb9a872404981a13f9a564d8ee83c : SUCCESS in 1h 07m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0cabfa93ec3d4913910361b6bcb2abe0 : FAILURE in 1h 33m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bffb6ced98a04e48b10a6cce991614e7 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":5},{"id":"889b5ce92e33fe0198525dce60697e37ca37b401","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-23 09:41:32.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/5/check/DellEMC_Unity_FC/9d12e49/DellEMC_Unity_FC/1932 : FAILURE in 12m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/5/check/DellEMC_Unity_iSCSI/c2f99fd/DellEMC_Unity_iSCSI/1960 : FAILURE in 2h 19m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"f7b036566ecfb06d3c7d845fb58d0bf054296d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-24 10:13:56.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"34e10a3e008969c5d62eba1986f6e0ece21062cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 10:26:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3e51a5cfed14923bd346b4ce58afdc3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/144370288832410a94ebbff70ec6e3fb : FAILURE in 10m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f7f8ce49bd74a459df2a5358747eb8f : SUCCESS in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7686b5d30330b9c219ce3b7730045adeb953e2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:32:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0e5d0d98bc840608a3e03abbb3dcff2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/208a939634d34b82bd572ed682dde020 : FAILURE in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/979932e73d614518a6f16209774a8838 : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8703cf782a714cbf9db9335d99437716 : FAILURE in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/775020e5c6e14b7a8dc07f4e6efa66e3 : SUCCESS in 17m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72171923b6bc47b1b58ead8da2435c73 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/adef9a7a78e14260affc1c555e3cff66 : SUCCESS in 1h 06m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c0f387651932452e8c463f246cd3cdcb : SUCCESS in 1h 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/355a2daa39b04e39a0451fd2a647f67c : SUCCESS in 1h 03m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb0789f8189d40cca5061d6d5218a1d7 : SUCCESS in 13m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d77fbfd3b274237b5152d8cc84e60d3 : SUCCESS in 19m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bb2cf030b71a4c11b44780809c6f4519 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a5c0f283a874e96956b35bb0d6b0184 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7197bb4813de45e68b05fa0a9bf4e045 : SUCCESS in 6m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5204f504abaa41e29ec0cb71e9eb969e : SUCCESS in 8m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8ade26b666b142aeb58fb50061d2677a : FAILURE in 46m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/137d88aae48e478ebb7337717f33d850 : FAILURE in 7m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/252b0d0e5efe4d1eb9144f22e920895a : SUCCESS in 2h 14m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9f05a9e927134c8fa996e2374d6bd986 : FAILURE in 2h 12m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c0581be6485c4fc39a2d4c8e98662354 : SUCCESS in 1h 28m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0391d7250dc84b1da12c7bbf71bcf917 : SUCCESS in 1h 42m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4073fba16b0b48d29d75653dd47e14a2 : SUCCESS in 42m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b53d89e7032643f0ad3ccca3b9cd61f0 : SUCCESS in 1h 15m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86d59ba745bd40759952e083b43d5d1c : SUCCESS in 1h 29m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4e47469da6a54c2fa6f7325aa0748dd9 : SUCCESS in 44m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57068aef7f3349a68149d0931c90ee70 : SUCCESS in 1h 20m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ef02241eeaae4114886de645d8e5371a : SUCCESS in 1h 24m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e0304002d35448bda237ecf6925519e3 : SUCCESS in 1h 19m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b3eadc6f6784869a02bc37817169874 : SUCCESS in 1h 06m 01s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/fujitsu/eternus_dx/eternus_dx_cli.py","accounts_in_message":[],"_revision_number":6},{"id":"c48a2751f84eabcc37b7c20eb2164f531e338eef","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-24 13:06:52.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/366f3395fcdc47898c73bf7bb3989891\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9da29034f3c74e27b2400b0e2e934f86 : POST_FAILURE in 23m 56s","accounts_in_message":[],"_revision_number":6},{"id":"f9aefda762a4826aadc2844a6a6388e950f6e0dd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-24 14:50:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 22m 53s","accounts_in_message":[],"_revision_number":6},{"id":"ed71b915f1983aceb838fe644bdcfe69e89cf38e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-24 16:54:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 35m 45s","accounts_in_message":[],"_revision_number":6},{"id":"200a36a78f1aa21e4f72cb799ea59220e19d8d5a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-25 03:46:12.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/495/.","accounts_in_message":[],"_revision_number":6},{"id":"0832ac0360efb07c930e60545a96023727e9802b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-25 22:42:29.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/04/913204/6/check/DellEMC_PowerFlex_3.6/95716b8/DellEMC_PowerFlex_3.6/1449 : SUCCESS in 1h 49m 45s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/04/913204/6/check/DellEMC_PowerFlex_4.5/b3950bc/DellEMC_PowerFlex_4.5/363 : FAILURE in 1h 58m 00s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/730f75f/DellEMC_PowerFlex_4.x_NVMeTCP/303 : FAILURE in 2h 56m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"80a6169d4ba0cbbc00846089385ea339681131b5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-26 01:57:19.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/6/check/DellEMC_Unity_FC/4f728a7/DellEMC_Unity_FC/1968 : FAILURE in 1h 21m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/6/check/DellEMC_Unity_iSCSI/7c29a1c/DellEMC_Unity_iSCSI/1993 : FAILURE in 1h 41m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"6a5c068605cb6c61f18bd3277c7d9e5a146b37b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-27 09:55:39.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"1fb123dd1d305624179de8e66af62ad2356fba71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 10:21:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38dedb8a01114e9a8f4224aa97723857\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/987322b46d924c6f8f704a725d23efa3 : FAILURE in 14m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6aa2e426168141e9a12232cae0983dd6 : SUCCESS in 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fc89d1d85997b85d9a00c9ed0bffa9b39bb33560","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-27 11:35:57.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/061ab089ec324fb28cef627503031499\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c194e592f11749c3b80d231068b6abfc : POST_FAILURE in 22m 21s","accounts_in_message":[],"_revision_number":7},{"id":"53b9e9c89ef0b7408c268912a9233faecaff6e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 12:30:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6117d323e3d049c79c3ef32223cd88fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/574937f5d51a41cb90608341485425cf : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97553cbda6194f889da4e5292993ba5d : SUCCESS in 10m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0dbfdb6f298b4d0c9f2e08da671dcd0a : SUCCESS in 11m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc715cac054442e8a664421b42596b5b : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810417cbca4f4ae3aa610e845d374d31 : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/c3d6817596af4b01a6d4ce6991bada67 : SUCCESS in 1h 05m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8eda6e87acd4dce85dc30e8013403a6 : SUCCESS in 1h 30m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3ef57bccf0245d1b63e2e041c7fc921 : SUCCESS in 1h 20m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b9527da7d0c4bbeb6b38cef1a304671 : SUCCESS in 17m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fd9b3f3ef6d14abb9c450f3de34c969e : FAILURE in 29m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1d997f9b1bb944e89443ada23f16fa6f : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01a8c18a3cc74f3a94abe70414d9eb2e : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/24519c8794b3438b963c90d59a80307b : SUCCESS in 7m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7ead6298ebde4b76a3935c93bce95847 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b24abc83ab2b416dbc4d5e105a9d77d0 : FAILURE in 40m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8343ce7a3a314250a66b45cb8bfbc338 : FAILURE in 7m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/962400b0d0a9400ea2e62eb584526e26 : SUCCESS in 1h 58m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/12b05541cc554e1bb52a64f96f9379c2 : SUCCESS in 2h 30m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/93c6d1f03c97441dba13d6976fffbbc7 : SUCCESS in 1h 39m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/62b1afe788554436ad667194ce1f0c5e : SUCCESS in 1h 53m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a87c21adda804b1cbabf345f4465427a : SUCCESS in 40m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/44756afc69964badb62a883afb32258a : SUCCESS in 1h 14m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b51415141ee64421b9b984e912249626 : SUCCESS in 1h 32m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f9b7fd1322b54436a2ed5a700c788470 : FAILURE in 48m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a021fda50d2946d3b0040f317a2dedb6 : SUCCESS in 1h 39m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ba299c1cdcbb45c08d3832115c930ac3 : SUCCESS in 1h 36m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9569298486645bc893ebdc3661c1fcf : SUCCESS in 1h 25m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65e9be70cfa84bd3ac1ea1028d414874 : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":7},{"id":"c00a01e137d41a9549f6767799c4d45e47bfaca1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-27 14:13:04.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/7/upstream-check/cinder-tempest-plugin-ontap-fcp/e9691a3/ : RETRY_LIMIT in 4m 08s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/8bf54cb/ : RETRY_LIMIT in 3m 54s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/04a7d0ddb30349ecab6c7ed2da6a8351 : RETRY_LIMIT in 2m 15s\n- cinder-tempest-plugin-ontap-nvme-tcp finger://managesf.services.openstack.netapp.com:7979/bfa70d64a6e04942be03389e6167f889 : RETRY_LIMIT in 2m 14s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"4f95af1553143d60387bea0958c77861f05cd9a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-27 16:55:04.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 36m 08s","accounts_in_message":[],"_revision_number":7},{"id":"b5ef439c396b830adb5d75063560e0925660dab6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-27 17:19:42.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_PowerStore_iSCSI/5d434ca/DellEMC_PowerStore_iSCSI/1387 : SUCCESS in 1h 30m 53s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_PowerStore_FC/9ce1704/DellEMC_PowerStore_FC/1570 : SUCCESS in 1h 19m 26s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_PowerStore_NFS/f02b831/DellEMC_PowerStore_NFS/1081 : SUCCESS in 1h 19m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_PowerStore_NVMeTCP/608892d/DellEMC_PowerStore_NVMeTCP/935 : SUCCESS in 2h 02m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"d0d4b4c59109436228071b8b8f1a123d103cc10c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-27 23:42:33.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/516/.","accounts_in_message":[],"_revision_number":7},{"id":"10fd093bd4e575163e44e4d75139f2e9f2feb731","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-28 09:42:30.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_Unity_FC/079beaa/DellEMC_Unity_FC/1989 : FAILURE in 2h 10m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_Unity_iSCSI/e666878/DellEMC_Unity_iSCSI/2014 : SUCCESS in 2h 30m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"84899f8fdad98ce3d394e34b7160670aeee04a35","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-05-28 15:17:45.000000000","message":"Patch Set 7:\n\nBuild failed.","accounts_in_message":[],"_revision_number":7},{"id":"8fed2fc0dbbb4cb5b11f1fe298ab71434fcdff33","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-05-28 15:52:22.000000000","message":"Patch Set 7:\n\nBuild failed.","accounts_in_message":[],"_revision_number":7},{"id":"63382084c13eab7f4175f52c2ec69012d6c87544","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-28 17:19:40.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_PowerFlex_3.6/3f94eef/DellEMC_PowerFlex_3.6/1470 : SUCCESS in 1h 54m 52s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_PowerFlex_4.5/e4e7055/DellEMC_PowerFlex_4.5/384 : FAILURE in 1h 57m 44s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/7eb56bb/DellEMC_PowerFlex_4.x_NVMeTCP/324 : FAILURE in 2h 55m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"661cb4c5981a8a3f63aff9fab40e0f84ba3058f1","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-05-29 02:05:33.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/913204-7/Seagate-iSCSI/ : SUCCESS in 2:26:04\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/913204-7/Seagate-FC/ : SUCCESS in 2:17:17","accounts_in_message":[],"_revision_number":7},{"id":"1f2359477bc2a922cbb9603506a69356bb199cdc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-30 14:00:07.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/7/fc : SUCCESS in 0h 44m 54s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/7/iscsi : SUCCESS in 0h 58m 00s","accounts_in_message":[],"_revision_number":7},{"id":"7b64f96e5ad6b1808007383e3467c23015960da0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-06-07 15:42:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/7/20240607161059/NEC-ISCSI : SUCCESS in 5h 04m 21s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/7/20240607161059/NEC-FC : SUCCESS in 8h 31m 54s","accounts_in_message":[],"_revision_number":7},{"id":"d23bff6783f607fdc614433da1f829c90844a54e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-06-08 00:04:49.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/7/20240608012030/NEC-V-ISCSI : SUCCESS in 3h 43m 53s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/7/20240608012030/NEC-V-FC : FAILURE in 7h 44m 17s","accounts_in_message":[],"_revision_number":7},{"id":"0ec0b98010e676040cdcc27c7d269fec44256904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-14 07:01:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b8a1963617bbf3e407cc8ffc17a229d1d7a9c6fa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-14 07:04:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"6f87a5a7ac2f99f004d6ad6e6c866a5b7e260419","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 07:25:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17a59bdfeaa04c39a3b207218943f326\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22ca01c7a4ed4253ad0c6a6edc86f683 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/445657c0ad0c405d9460322d42cd2475 : FAILURE in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2e272628b15970cbb54df8b2672836f0e3d11668","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-14 09:24:24.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/965e647cd2e245d9a193e36fb18ee296\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/17d20ec3dc684c8ab904e109d2e66c5f : POST_FAILURE in 27m 09s","accounts_in_message":[],"_revision_number":8},{"id":"6e8fb1e6575e0acbbbfc5b89753752930be72e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 09:54:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/368ced19606d4088af6b91f9da19873a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092efc1538b147b0a9884d94ffbae1b9 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/622b74cded8e4aa7aded4c715745f5c8 : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db25b787252442de98ae9e72aafd5b80 : SUCCESS in 14m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ba97637491348b8bb783e41f4ce5972 : SUCCESS in 15m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7bdb287fa034837bed6b27b564a7ca4 : SUCCESS in 11m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/2f92a897dbb7466386f8cf6a9356d57a : SUCCESS in 1h 01m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e59782ff27d14995bc251b7f1d88330e : SUCCESS in 1h 32m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a94be85f18a646adb1354fc68b9ec252 : SUCCESS in 1h 02m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e072e154640142e4b787f1cc875e27a8 : SUCCESS in 16m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/175a185e7ba448bf8799c52ea5289e04 : FAILURE in 13m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6d761a7309f64aa3a903d5643120ed0a : SUCCESS in 7m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e85cbc340d349249c6747e07cd6a730 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc8141c50e774f55a502044bbebe70f3 : SUCCESS in 7m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c2f07ff54e154b9882d630cd342381dc : SUCCESS in 7m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6687d98868ae4f4a98794c8e6ca5fbb3 : FAILURE in 48m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c46eb097f0bf43b7b58692ce40d5ef3d : FAILURE in 7m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f9ffb5a4465445887d4b573a59817dc : SUCCESS in 2h 38m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7bb8e0494ca44df38453873652ec83cc : SUCCESS in 2h 48m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/63184bbe9a924b30b4e1be1e0fe1fc49 : SUCCESS in 1h 37m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2ca15a93a6d643df960b2a7342e89ad2 : SUCCESS in 2h 10m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ff5544d42e1f4a8ab0f0107fbc964547 : SUCCESS in 33m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0584d0aa1dc84c03a115d8c7061ce432 : SUCCESS in 51m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9d0b08cc603a4dd99bb2cbb3e9540f2c : SUCCESS in 1h 56m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/060a0f38a1c840159026f32024ea3436 : SUCCESS in 58m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4ee762cae0c34170abad516620e2547a : SUCCESS in 1h 36m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f402b69af52647598819f6d9bc8329c5 : SUCCESS in 1h 29m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4750a251846b45e998dfd77c11ba4c8d : SUCCESS in 2h 00m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e97bf0b23f6e4d95ba6dc4b650f4169a : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":8},{"id":"b10ad21ce9cfe6736a63767f65d15ceedf1b7d48","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-06-14 10:00:37.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_06_2024_09_54_36\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"b6ea5e0ed165ebd10a033bf515b167f5541628f7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-14 11:30:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/8aaaa89/ : SUCCESS in 52m 04s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/8/upstream-check/cinder-tempest-plugin-ontap-nfs/8b7d0ef/ : SUCCESS in 49m 30s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"f8266b699c18b1b4c2359d13d07598ca61da8c2a","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-14 15:34:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913204 : FAILURE in 32m 16s","accounts_in_message":[],"_revision_number":8},{"id":"9ddfddf1898665b933a5ece63a3f031e4a3f478d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-14 16:56:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 34m 50s","accounts_in_message":[],"_revision_number":8},{"id":"d8d06eff5e16a41304a3dff9050a4a1948733dee","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-14 23:49:01.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_PowerStore_iSCSI/fcd254d/DellEMC_PowerStore_iSCSI/1442 : SUCCESS in 1h 24m 08s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_PowerStore_FC/3dd341b/DellEMC_PowerStore_FC/1625 : SUCCESS in 1h 06m 38s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_PowerStore_NFS/6c455f3/DellEMC_PowerStore_NFS/1136 : SUCCESS in 1h 09m 05s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_PowerStore_NVMeTCP/a65b538/DellEMC_PowerStore_NVMeTCP/989 : SUCCESS in 1h 58m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"d9f59ef27242dccced5f616e77dc8ed786f21719","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-15 01:35:10.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-8/FC/ : SUCCESS in 5h 37m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-8/iSCSI/ : FAILURE in 8h 31m 12s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"fce6faad9ed5907f04537b6208638ac2e6599aef","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-15 05:29:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/8/check/fujitsu-eternus-dx-iscsi/acefcfe : FAILURE in 2h 26m 23s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/8/check/fujitsu-eternus-dx-fc/b0cfc7b : FAILURE in 3h 54m 03s","accounts_in_message":[],"_revision_number":8},{"id":"c8f4ff99c4b5f2bba1c54088a845a7c585af3210","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 05:43:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/119/.","accounts_in_message":[],"_revision_number":8},{"id":"3ab48bfc071da066740e7d0ce39286f296d7b0e0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-15 08:15:32.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/731893a077a6409aa47917298fc28cd9 : SUCCESS in 1h 16m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/18a6d58f8e124fbea06db4e488373087 : SUCCESS in 1h 18m 47s","accounts_in_message":[],"_revision_number":8},{"id":"faa085114ee0684bb75a7c863ad93f22d4d27786","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 08:50:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/605/.","accounts_in_message":[],"_revision_number":8},{"id":"e6b9ca5f0ab9021b1376bae7d59119c8a420fe83","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-15 18:44:46.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_PowerFlex_3.6/35fc430/DellEMC_PowerFlex_3.6/1580 : SUCCESS in 1h 50m 27s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_PowerFlex_4.5/6779f22/DellEMC_PowerFlex_4.5/475 : FAILURE in 38m 30s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/19715b2/DellEMC_PowerFlex_4.x_NVMeTCP/425 : FAILURE in 11m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"3cb7ef6935ec14a375182acfb7f97e54d40a1e5c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-16 19:18:54.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_Unity_FC/29a327d/DellEMC_Unity_FC/2089 : SUCCESS in 1h 33m 57s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/8/check/DellEMC_Unity_iSCSI/4bbc559/DellEMC_Unity_iSCSI/2112 : ABORTED in 6h 00m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"bbbd68b0851fc9017bb2396ebf5cc8ec1d14f6e4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-06-22 21:04:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/913204-8/HPMSA-iSCSI/ : SUCCESS in 3:16:07\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/913204-8/HPMSA-FC/ : SUCCESS in 2:29:40","accounts_in_message":[],"_revision_number":8},{"id":"8dc520494f3c937284437997825966cd03c91d89","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-06-23 18:35:42.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/913204-8/Seagate-iSCSI/ : SUCCESS in 2:20:19\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/913204-8/Seagate-FC/ : SUCCESS in 2:12:52","accounts_in_message":[],"_revision_number":8},{"id":"b9c0e9755efbcec759e0dd46d630e64517f8d304","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-06 02:41:51.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/8/20240705193033/NEC-V-FC : FAILURE in 7h 31m 03s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/8/20240705193033/NEC-V-ISCSI : FAILURE in 16h 11m 16s","accounts_in_message":[],"_revision_number":8},{"id":"8cc946334c123c1fde5b659ecf950bfe4e514c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-12 02:26:09.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2a8810c8de3dd19086f2ea990c106a687b1782b3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-12 02:41:50.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":9},{"id":"cab5d68c64e4673f6618151f3742f64b8b3a9036","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-12 02:42:39.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913204 : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":9},{"id":"71021ffa6bd2d7bd9a5e0640abe5f3db563ebcdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 02:51:25.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6ddbf5bc9854e289320b4594c79637b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6814e0fb3e634e009f233b890bc6365b : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1b7a7b72a16949169ce1da0b1f7bd2ba : SUCCESS in 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cc9ba08031f4329962e31cbb2db6308bd3c33900","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-12 02:55:33.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8501a5c6d8bf4f2b9f77ff6029f226c5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6e6f219c6e1648e4bd70227e87c1f591 : POST_FAILURE in 22m 51s","accounts_in_message":[],"_revision_number":9},{"id":"70cfa17e08a88880b910375b1145adfc1547b79d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-12 04:26:59.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ca551b52a48246ed9230857fc3ee0da1 : SUCCESS in 1h 20m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cc2a86441ca34ce3844c1f4d9e24d505 : FAILURE in 1h 09m 13s","accounts_in_message":[],"_revision_number":9},{"id":"1aa25742790dcbc8954823ebaea8c5cf58a4a62f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 04:48:01.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2967a67962034c35b50c965ad2068992\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ab942274734dab9cc26ba6ad32f11b : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bf38b5d467c42fd8b1a307654320b82 : SUCCESS in 10m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dda8ddb39d984d1b8e3c2f0fe02c8d8d : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f7eb9b6abe14942b76570b9a11e8a69 : SUCCESS in 15m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bbf4d4dab74d19b32e0ef911b56145 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/bb951552ee214b4b9d1a6ca684c78931 : SUCCESS in 55m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32b07781a0464faf897c050c2e05cc71 : SUCCESS in 1h 31m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/766e1a200c1a4fe681d7402ee0ef8a9e : SUCCESS in 55m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11dd7a78b3634cd8aa4a7e2e153b9207 : SUCCESS in 18m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3074c46aab2e41f386fc45d4c25f7d68 : SUCCESS in 17m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3398be3521e441babb5c4ab13b969db0 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ae0f2ad4903745ea813aafd9e163ea65 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7fd776d07a0486ab414eb718bfd5f0e : SUCCESS in 7m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/300e9ac8879a4fa98b1b73b0c00d6efd : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9f4b4d362e24e21bd92e1080e9200c5 : FAILURE in 31m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/469b8ccc4c4942a2a40324d65cdf53c2 : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/77253d7548724f5f98af713ce3ff06b2 : FAILURE in 1h 38m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9f5aa9032ed0477a8a529ff3f8bd76c0 : SUCCESS in 2h 09m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db9d96d478fd459c8b2fd00100aa78d5 : SUCCESS in 1h 57m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/38743d341ea540309e00b4d20c7c629b : SUCCESS in 1h 54m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/943cbc76241e4f698422154efd1db0cc : SUCCESS in 36m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b01f954e1034d219c94123a1701dfd6 : SUCCESS in 1h 17m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9a683390559845c78387e665b495fe07 : SUCCESS in 1h 40m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3733f7ec9c8c4933a5d075f5898e8491 : SUCCESS in 53m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4815c922edf74f3692f0b3adb73b79e6 : SUCCESS in 1h 30m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a39f5a5afc91420fa17cdd45eeaba4cb : SUCCESS in 1h 40m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b70178f04c3442e29d6b665a542661bc : SUCCESS in 1h 27m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80a44c8de8f04c91aa60894a171a0b4e : SUCCESS in 47m 27s","accounts_in_message":[],"_revision_number":9},{"id":"e35d5966860376449fb7310b6290f9412aafc108","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-12 04:50:06.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/714/.","accounts_in_message":[],"_revision_number":9},{"id":"4544a6a028ed4171396be5d1b4d4e9554ccdce83","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-12 05:22:35.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/9/check/DellEMC_Unity_FC/f8da965/DellEMC_Unity_FC/2207 : SUCCESS in 1h 39m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/9/check/DellEMC_Unity_iSCSI/6e0de66/DellEMC_Unity_iSCSI/2216 : SUCCESS in 1h 43m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"710b43d65c9c30ce91919326fff1266df51dcd14","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-12 05:29:21.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-9/FC/ : SUCCESS in 3h 2m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-9/iSCSI/ : SUCCESS in 1h 52m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"7b5cf53d149e14237198d84a970aa3c38a8e8a07","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-12 05:34:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/217/.","accounts_in_message":[],"_revision_number":9},{"id":"960d0ab3b70e528e0be832d8a72fae553b14f116","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-12 05:49:37.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/04/913204/9/check/DellEMC_PowerFlex_3.6/4127e2a/DellEMC_PowerFlex_3.6/1692 : SUCCESS in 1h 52m 50s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/04/913204/9/check/DellEMC_PowerFlex_4.5/a499998/DellEMC_PowerFlex_4.5/573 : SUCCESS in 1h 51m 54s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/8e0c195/DellEMC_PowerFlex_4.x_NVMeTCP/527 : FAILURE in 13m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"b88ca769b04645753cc1d17bb031e79022835369","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-12 06:22:11.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/9/check/fujitsu-eternus-dx-iscsi/639f959 : FAILURE in 3h 50m 23s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/9/check/fujitsu-eternus-dx-fc/5d0f51e : SUCCESS in 2h 43m 43s","accounts_in_message":[],"_revision_number":9},{"id":"66788892852e00608e2688c0e90b8470b7216d8f","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-12 06:36:47.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"af97df01495726125084106ccdb392f4695fb6fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 06:41:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d3b454418a4d52ad1830aa4cb8ae1b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09f1440806d44ad7983cc74ee472f3a1 : FAILURE in 3m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ba800c689324c48abdf2ace13a7b556 : FAILURE in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c8fbe2d8b846ae1aa67e2a12c59f6b2d952a1283","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-12 07:12:36.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/693696a733d24904a675a7ae52afce21\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ef8b0b72b08040aab63dccce7d0adccd : POST_FAILURE in 22m 58s","accounts_in_message":[],"_revision_number":9},{"id":"706bc8cc5bc2f2fa73b9b6999dc4635e3e47ff33","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-12 07:59:34.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/53ea60267f1c4210971422cd7eac891b : FAILURE in 1h 17m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/600022aabc1c407abe2af75a4e1e4be7 : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":9},{"id":"8c28afca6a4d6f12eab3cd6ff69a79e6d5c1553b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-12 09:47:58.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6fb197823108807269a57831f6e77edec186bbcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:53:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63c838c632db4879bfaf85b5eff204da\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dac5789b9d1b4013ad91c050f660cfcb : FAILURE in 3m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f3e64cf4fa84154a6d95119a7f43603 : FAILURE in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5f12c40d46a99b11ce94a7eea16d8d5df0b82c53","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-12 10:19:37.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913204 : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":10},{"id":"bc6ae2f0fba539a5bfb87b08b91cef2be4efecec","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-12 10:20:04.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":10},{"id":"c954b68e1327429907b7a0ff42f28d6350644f64","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-12 10:47:25.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b69509f9f2a54e019c198a7b6ab253c2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2ad19a7240424d0cbf117c972bc3c547 : POST_FAILURE in 22m 58s","accounts_in_message":[],"_revision_number":10},{"id":"151c90ccf43e7a381ae8a48db44b4885f32d178d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:07:25.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f1d3a5374014b02955939baea8b1f57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f90928c35b049669d17d85ae7a74b46 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39b582921eaf486199c167f49fc80271 : SUCCESS in 10m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/354e2d7c29b748b88879fb816288335a : SUCCESS in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8eeb538878c54e0e88bc789299fad09b : SUCCESS in 16m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe0006817db4b83a8a4821950eba03d : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/e04c4b8cf6144aba9f2a3e55497d3303 : SUCCESS in 1h 10m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e79e166d8ee8475aba6567e0d8b65ee9 : SUCCESS in 1h 35m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ddff45fbb144f91a6258a7a7917bb05 : SUCCESS in 1h 00m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/915ab01f28b64eb4b0f079e1edfe09b0 : SUCCESS in 20m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6238103f6f3f45e0a1214fdd38b89d0c : SUCCESS in 18m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b70308a4cba141d0936e8f6783683981 : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3aafe735bab941a39d84be1a1762d645 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bdbb878d166a4f90b842a6254c07825c : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a29dc92a6ba34c24bdcdc92b1bd69249 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9baa1bff94694722b77fdb1b1c128afe : FAILURE in 38m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ade937b2c91401391abadfc8578e604 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a65b3f502cb94d2a954736ed8ee7db7f : SUCCESS in 1h 51m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a9d0172b21364f4d948c12ec2308aa8a : SUCCESS in 2h 11m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c037526ae4e447bbadafa2a9ad79082 : SUCCESS in 1h 53m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8121a2a34f1643b9ab2f0cb4899308bb : SUCCESS in 1h 44m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/66c427547b8b4435ad369915f9c27319 : SUCCESS in 40m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5e8474ff74054ab3ba6ce1ce8a4015ae : SUCCESS in 1h 14m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1f19cce659534600a67d779e20f5240e : SUCCESS in 1h 34m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8f5182363564471f9f179f9d4e4ca958 : SUCCESS in 42m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/06b325f12d64477ca4825fdf4ee67fd6 : SUCCESS in 1h 21m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5754b0bc383c438ba1ee484ff784250d : SUCCESS in 1h 39m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/83fbe09036a5444a8933064fbd6a4dee : SUCCESS in 1h 46m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc51e61fef2f4753884b2a84bce01ef4 : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":10},{"id":"81ff7d72262fb35417431a844aaf2ad89a726afc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-12 12:37:14.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d5dc59100ffd4c9aa36b73d26ecb4a75 : SUCCESS in 1h 18m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/15001c2ff11a405dac0c55de4a6cf178 : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":10},{"id":"d96a0466fcf01f0564259709642802da1c785eed","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-12 12:46:27.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-iscsi/5fc3530 : SUCCESS in 2h 47m 19s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-fc/d5c1922 : FAILURE in 4m 01s","accounts_in_message":[],"_revision_number":10},{"id":"be99e9d452a59562ba2270b8f66e1c5318f3ac4a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-12 13:33:01.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_Unity_FC/20df0cb/DellEMC_Unity_FC/2214 : SUCCESS in 1h 25m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_Unity_iSCSI/b29f712/DellEMC_Unity_iSCSI/2223 : SUCCESS in 1h 46m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"d693be649d1f936a205f20329392817105378c80","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-12 13:40:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/c8d8922/ : RETRY_LIMIT in 25m 59s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/10/upstream-check/cinder-tempest-plugin-ontap-nfs/5b35461/ : RETRY_LIMIT in 14m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"26af769b4b1e787cb708bfec85950e9f4b877ed9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-12 13:49:04.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_PowerFlex_3.6/15bf2c8/DellEMC_PowerFlex_3.6/1699 : SUCCESS in 1h 58m 47s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_PowerFlex_4.5/7ad90d0/DellEMC_PowerFlex_4.5/580 : SUCCESS in 2h 11m 57s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/a0e8a64/DellEMC_PowerFlex_4.x_NVMeTCP/534 : FAILURE in 12m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"af094f62e0484037e59927de60f6f88895307d28","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-12 14:17:50.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/720/.","accounts_in_message":[],"_revision_number":10},{"id":"286fb40288ae9b50f04d8e2391804a64c7f9f30f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-12 15:00:25.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_PowerStore_iSCSI/6b78ea9/DellEMC_PowerStore_iSCSI/1511 : SUCCESS in 1h 19m 46s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_PowerStore_FC/58f296c/DellEMC_PowerStore_FC/1708 : SUCCESS in 1h 12m 52s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_PowerStore_NFS/490edf9/DellEMC_PowerStore_NFS/1206 : FAILURE in 1h 11m 27s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_PowerStore_NVMeTCP/b6e8007/DellEMC_PowerStore_NVMeTCP/1058 : SUCCESS in 1h 48m 21s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/04/913204/10/check/DellEMC_PowerStore_NVMeFC/6236b97/DellEMC_PowerStore_NVMeFC/23 : FAILURE in 2h 51m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"cb80bfe271d97830e7f3ae8b2427909f17d19b93","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-12 15:36:50.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/223/.","accounts_in_message":[],"_revision_number":10},{"id":"55359f66917666501b032b315c1599caaffb381f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-12 17:37:21.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/10/20240712205235/NEC-ISCSI : SUCCESS in 5h 18m 52s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/10/20240712205235/NEC-FC : SUCCESS in 5h 44m 42s","accounts_in_message":[],"_revision_number":10},{"id":"a3e050742cb7dddda9acb56b8ad1aba69f963f2f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-12 21:51:44.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/10/20240713030516/NEC-V-ISCSI : SUCCESS in 3h 33m 19s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/10/20240713030516/NEC-V-FC : SUCCESS in 3h 46m 24s","accounts_in_message":[],"_revision_number":10},{"id":"23839f8da67af86245323a7b5ebf39490ec25d62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-14 10:37:46.000000000","message":"Patch Set 10:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/14_07_2024_10_32_25\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"4c3c64aec04d88bbced29ec9679780d9162cdff1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-14 12:13:10.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_07_2024_12_07_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"9f0c499e8d8d7c3f70c2089181bbf45d8357f7f7","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-16 05:36:18.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d757ad859c58f7d0b151247dfffaa5a7c85ee9ce","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-16 08:07:52.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-iscsi/d6ff9b3 : FAILURE in 2h 23m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-fc/6ff3c6d : FAILURE in 2h 25m 47s","accounts_in_message":[],"_revision_number":10},{"id":"37ae47c591562f8fc9b249d61afec45d0ff84166","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-18 02:35:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fe64a3a7a6de15f57eb36d2c434dce3682b298e7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-18 05:11:52.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-iscsi/b31abff : FAILURE in 2h 30m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-fc/a79ec09 : FAILURE in 17s","accounts_in_message":[],"_revision_number":10},{"id":"4dfe204f267f4fe361570c1929f45a6b249097d0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-18 05:23:20.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ab7c6e880d96e68e1f94e5db1babac313682331b","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-19 03:00:11.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a563a4788904379a0ab2f525f51a7129879df3e6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-19 06:57:05.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-iscsi/0546377 : FAILURE in 1h 59m 16s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-fc/0d150f5 : FAILURE in 3h 51m 23s","accounts_in_message":[],"_revision_number":10},{"id":"db1164bcc946f864430957fb463659cdef7ad9a4","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-22 03:49:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1abd74edab5ff573d316c7da1460b92309684522","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-22 05:30:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-iscsi/70d25c1 : FAILURE in 1h 35m 37s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-fc/7b7a24a : FAILURE in 1h 34m 26s","accounts_in_message":[],"_revision_number":10},{"id":"5493bfde8a4b7f1446d1fd2e02275c9e4f871100","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-22 05:43:13.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b7512a9573085c6193d1d552f1f5a0628e578654","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-22 07:47:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-iscsi/e4302f6 : FAILURE in 1h 37m 16s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-fc/c76b2ae : FAILURE in 1h 26m 54s","accounts_in_message":[],"_revision_number":10},{"id":"21b837d75142801946f7c9bc32d32330db4020f6","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-22 07:48:52.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b9aa972f50384e0830401f0b5160584fba81b25a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-22 09:02:50.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-iscsi/30003c4 : ABORTED in 1h 08m 53s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-fc/83b202b : ABORTED in 1h 08m 46s","accounts_in_message":[],"_revision_number":10},{"id":"459dcdcf49e76847f8707ed1d08cb8456f7e5aca","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-22 09:47:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"44697d00e450aee0dfd50b63baf540f3c9ad61a3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-22 12:03:32.000000000","message":"Patch Set 10:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-iscsi/8162bae : SUCCESS in 2h 11m 15s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/10/check/fujitsu-eternus-dx-fc/dece6f2 : SUCCESS in 2h 08m 44s","accounts_in_message":[],"_revision_number":10},{"id":"eae6013fafccc07f9af9491bb48fb471a933cfca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-06 06:48:17.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"5d16527a55ba1c62936559b0f2fcd598750df138","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-06 06:48:28.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/223/","accounts_in_message":[],"_revision_number":11},{"id":"f1161abca62b50589ff4a45d70a251bf5e03f6bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-06 06:48:29.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/223/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"548ef53838f15eacdedfd9b4aa47381243e7825e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-06 06:50:15.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"26ece84e4f4b067b78ae0da6a8d4592df2c912ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 07:12:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a850e416b3c043caa29c0e61215c4c29\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42e783ae64c84800961f5ed2cffa0a66 : FAILURE in 21m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bda28285cab4d10b1e0680ba2db6fe5 : SUCCESS in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"80124f5b401f94b9838983e2edfe1ff0e80cefba","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-06 07:37:46.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/625ca572f9124de988b47be62052a19c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0028ab48c86c4b54973adf760c597ba7 : POST_FAILURE in 27m 18s","accounts_in_message":[],"_revision_number":11},{"id":"f4be5551bbb3c3b05d3e78fead4e8c531a4dde8c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-06 08:02:01.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 34m 30s","accounts_in_message":[],"_revision_number":11},{"id":"7e68575a90347a132e402649a39399e85aa9e6df","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-06 08:16:06.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 33m 57s","accounts_in_message":[],"_revision_number":11},{"id":"f2cff69c367b61e79fcce318019400fcb6d1cc3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 08:49:59.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8568305d56c948e9aa439b84f174131c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a9d018bf55437e95848444527b4558 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1496f1af379b411f824b00cde3e56474 : TIMED_OUT in 1h 00m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c78b12d0f7db417592ca88eeb0fa8c96 : FAILURE in 15m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9202799ed31f49ab9edc80f8864a6c95 : FAILURE in 23m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d67a667328ff4b4fa4ec457a96dcc010 : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/8bacb3accb4b41d78762353cacec677f : SUCCESS in 1h 01m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03701b84065d45dfb0c13d9385d1b1ed : SUCCESS in 1h 22m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e34944b939cf4eec9a5c798cc0ec6581 : SUCCESS in 1h 00m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2105ab54524d45ef99fa57db5588bbe2 : SUCCESS in 19m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/27d61df23dbf494e8707cb324fb8b3d0 : SUCCESS in 16m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5c89c330b22a4ebf9a5d1322ce03037d : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2692173b6c554ef882349929022fc2f5 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/004e4e4bde5e40a0a11269f6bf3e0455 : SUCCESS in 6m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c81f5c9bc3b64f2681f6dc91fce8650b : SUCCESS in 8m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7dc48fdf5f914697a1388542b7858868 : FAILURE in 24m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f0af270fd084b4bb629e15860a19e6d : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7c9a6849cd114da2aea6c42c3457cff3 : SUCCESS in 1h 42m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c53899be68f94790bbdfb97a47a377b4 : SUCCESS in 1h 56m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/053f6f2e1fdc4ef999d90fb2e7d2e671 : SUCCESS in 1h 44m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cbb5bbe5e8514cccbcaf1d8802440cd3 : SUCCESS in 1h 38m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1019248cbeb642dd94beb68c0cb6be8c : SUCCESS in 35m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ff67fef7b55f4c549e747616a2249491 : SUCCESS in 1h 14m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/03e3705cb5fc43459391d4b631ceaad9 : SUCCESS in 1h 17m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/020258616a864106b507b21c7bd248db : SUCCESS in 42m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2b7a7a4986ed4e85aa22d68b6d24ef8e : SUCCESS in 1h 19m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f95710bc55b947bc9c0dcc7996d31d32 : SUCCESS in 1h 18m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a0252a7dfb5431f994f7fbf5d38c17f : SUCCESS in 1h 52m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cd52b8428504e798a005ee709c326fa : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":11},{"id":"d407672f2ec14e399f680c788c38fc65d65030b1","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-06 08:52:15.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"58538029c1a72706b6e3ccf0de07302a5169e13b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 09:21:16.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec536573be2e40b09c404b6b146c2f8a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/613a4f18b1b64c6cacf9e9731f258586 : FAILURE in 23m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d2f36a0d83bf44bb839a3252aa1c9f61 : FAILURE in 26m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e7d3cf9570b64daacfea42f02925d049d301a95d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-06 09:35:34.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1a6689141c394705b1113b8cddbb6aae\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/11950d08c07442548f55fb496800214a : POST_FAILURE in 27m 10s","accounts_in_message":[],"_revision_number":11},{"id":"69c7dc12200681ba5d4ae82dc7037674ecc46b2d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-06 09:52:27.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/37f469ab792d43dea81c8c8259cffb4d : FAILURE in 1h 23m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/01605ed7e2fb4f9c99ecc1a89c1981c9 : SUCCESS in 1h 12m 53s","accounts_in_message":[],"_revision_number":11},{"id":"583956bd6b0a6f2f1e336723e5328625249b9e89","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-06 10:59:22.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/849/.","accounts_in_message":[],"_revision_number":11},{"id":"c0c10f33b7fac2560f36a4e6955449b40ea94eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 11:31:32.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0460c7a1fd8a406dabf8894f15bf6796\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5e59dafb11426cb9a8a8499f24e93f : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2a1c765e4b84b6fa5f5c6d5d9d0392c : SUCCESS in 14m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5acaf19feb7c4dda9301a3ac33090b90 : SUCCESS in 10m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9da1117592c04c76ae2007e8e534bf78 : SUCCESS in 15m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43818492032e48f9af4adcddb04cd260 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/3b3631af5d3a41e39182e60df91599e8 : SUCCESS in 1h 08m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a36102060744c5395f151092e8443a5 : SUCCESS in 1h 15m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a3860d694304ad5a0cb6d590b4450f3 : SUCCESS in 45m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3207750605ee49cb9a7ebbc696941983 : SUCCESS in 19m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d26ae9f8f38545bcb3f4afa147aef986 : FAILURE in 12m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6c9f907fc92c41aaae5ff083eae281b7 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8b8fefddf0749c88d359d5ca682b7ce : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db908cdeda8a46b6b7f230ba41d7d8f5 : SUCCESS in 6m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dbb06887c61f4b68a2f32ee379f757e6 : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8971e2de85a9469c93e29c9677b6eb77 : FAILURE in 26m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f38a43af583c41a2a7d9ffd097927e78 : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/035f0eb895f14fc68c410900cf4de1c3 : SUCCESS in 1h 42m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/15554b89d179432f8bb7c80e91bbc3be : SUCCESS in 2h 28m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8095fcd48dc24298b134ae6865f14dc2 : SUCCESS in 1h 38m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/77f4bf79d9b14e658c2f9846b503ef45 : SUCCESS in 1h 35m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e48cc08dcfdf46d9a2fcaac27e5bb5fd : SUCCESS in 38m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79ca02e3fe0e4c6c8cd81bcf0856f294 : SUCCESS in 1h 02m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/15bfa1311b9541369e83c88a50adaa25 : SUCCESS in 1h 22m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3552eb23062d408cb436f051b7a7354a : SUCCESS in 49m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/311088d0e3914cf09285bd5752342327 : SUCCESS in 1h 17m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b57621c58e7f494481cad257c194ddf9 : POST_FAILURE in 2h 12m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/012dbe6369144ea68945b2d74976618c : SUCCESS in 1h 58m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/688c0dfd7d9544c9b511619898a7e7e4 : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":11},{"id":"da350370194e638a4b423008ceeb470b6e14b09b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-06 11:37:19.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_08_2024_11_31_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"dc18acaab569775a92aa4fab99db853d98939fa2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-06 11:46:33.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_Unity_FC/9385a25/DellEMC_Unity_FC/2362 : SUCCESS in 1h 39m 40s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_Unity_iSCSI/100e55f/DellEMC_Unity_iSCSI/2367 : SUCCESS in 1h 58m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"b10a3870b4e1464154c74184d466a80678466c3f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-06 12:03:34.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/11/upstream-check/cinder-tempest-plugin-ontap-iscsi/90f3386/ : RETRY_LIMIT in 8m 41s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/11/upstream-check/cinder-tempest-plugin-ontap-nfs/dd9333b/ : RETRY_LIMIT in 13m 02s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"b57cc9c72bb2a00bff89d3c57ecdef1abaa0a0e0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-06 12:40:12.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_PowerFlex_3.6/34e3d2e/DellEMC_PowerFlex_3.6/1848 : SUCCESS in 1h 59m 19s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_PowerFlex_4.5/a0306bc/DellEMC_PowerFlex_4.5/727 : SUCCESS in 1h 59m 39s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_PowerFlex_4.x_NVMeTCP/fca4c91/DellEMC_PowerFlex_4.x_NVMeTCP/692 : FAILURE in 13m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"0ae7f8ae15a08967dce257da84f4306844f9190d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-06 13:39:34.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/11/check/fujitsu-eternus-dx-iscsi/d247017 : FAILURE in 3h 50m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/11/check/fujitsu-eternus-dx-fc/d06ea72 : FAILURE in 2h 46m 57s","accounts_in_message":[],"_revision_number":11},{"id":"5493a6b53848c1ec3b2a90e4bc95c4a6e0489a75","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-06 15:49:05.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_PowerStore_iSCSI/3ce9e1f/DellEMC_PowerStore_iSCSI/1575 : SUCCESS in 1h 32m 45s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_PowerStore_FC/748f15b/DellEMC_PowerStore_FC/1770 : SUCCESS in 1h 11m 31s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_PowerStore_NFS/c90763b/DellEMC_PowerStore_NFS/1268 : FAILURE in 1h 21m 04s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_PowerStore_NVMeTCP/1573ccb/DellEMC_PowerStore_NVMeTCP/1127 : SUCCESS in 1h 56m 18s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/04/913204/11/check/DellEMC_PowerStore_NVMeFC/1102ff1/DellEMC_PowerStore_NVMeFC/91 : SUCCESS in 1h 09m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"db13f8da4a96a2db36ea9f711e5f3230423b1ce8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-06 23:27:30.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/11/20240807044249/NEC-V-ISCSI : SUCCESS in 3h 38m 16s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/11/20240807044249/NEC-V-FC : SUCCESS in 3h 44m 36s","accounts_in_message":[],"_revision_number":11},{"id":"3620b51798aa0500262dee2c23e00beecee68cf4","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-07 06:46:23.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"eb9cf0ea271b56ec05006d35b2d8702f5dc5cc88","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-07 13:12:41.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/11/20240807140132/NEC-FC : SUCCESS in 5h 44m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/913204/11/20240807140132/NEC-ISCSI : SUCCESS in 8h 11m 06s","accounts_in_message":[],"_revision_number":11},{"id":"4bfe346a09569b07ab8bf5650ca0e0f2f83c584f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-07 13:22:21.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/11/check/fujitsu-eternus-dx-iscsi/4d6c2e3 : FAILURE in 3h 50m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/11/check/fujitsu-eternus-dx-fc/629fb9f : FAILURE in 3h 43m 23s","accounts_in_message":[],"_revision_number":11},{"id":"0a13643c1b527574e23aa8b5667bddf29f378fde","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-09 01:21:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"391f488d84a5e9c9517fe509360cca78933a0c11","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-09 09:07:35.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/11/check/fujitsu-eternus-dx-iscsi/0e85238 : FAILURE in 3h 51m 11s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/11/check/fujitsu-eternus-dx-fc/66d78fa : FAILURE in 3h 50m 37s","accounts_in_message":[],"_revision_number":11},{"id":"9f7de35ea4c9ef3d996cc5d0a1bb3bb93c73012a","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-09 09:14:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4be0a1ac6abb2d25c86bc1167d45a687aaf92f0c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-09 11:48:40.000000000","message":"Patch Set 11:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/11/check/fujitsu-eternus-dx-iscsi/8c2c99d : SUCCESS in 2h 16m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/11/check/fujitsu-eternus-dx-fc/adfbd17 : SUCCESS in 2h 08m 00s","accounts_in_message":[],"_revision_number":11},{"id":"452d3b213c035b19b62f63a715f335a99f563d34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-09-02 03:15:18.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"7a5a310ba2134827cf0fc26e33f8982f6675960d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-02 03:15:32.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/453/","accounts_in_message":[],"_revision_number":12},{"id":"2002bc43bdff733d5d3f74775d3f1859cdae1505","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-02 03:20:43.000000000","message":"Patch Set 12:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913204/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913204/12/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"252b04e00778dfe677551a0a8bd8d81f57d5e255","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 03:34:32.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf514ca3bf740338ebc1d13b4883248\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b59d2c3bf74241229cd809ad9977ceb5 : SUCCESS in 16m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad0ebc8003224d92b28e1c3916f4c342 : SUCCESS in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8f6660e272c40f92201e644944639f2a0662dafa","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-02 03:52:49.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913204 : FAILURE in 33m 51s","accounts_in_message":[],"_revision_number":12},{"id":"520f49aeac6b5383ccf2fae652a5402397354061","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-02 03:57:56.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 36m 06s","accounts_in_message":[],"_revision_number":12},{"id":"317681f257d2ae6eb4f8dd609d9acf0aff79be7f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-02 04:00:43.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":12},{"id":"dfcc6ab897b489e14990e4d38e1296e2a689482a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-02 04:48:42.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/073ab43c4eb64dd183686b737a591f7b\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/524b51d051594086a540a26516fcf52c : SUCCESS in 51m 49s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b82709cfd0b948539abbc10de29e7064 : SUCCESS in 37m 40s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b86d3bc5dc5c4e55bafe030b74eb4b8a : FAILURE in 1h 31m 53s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/20493df2497e40398d2737348bd986d8 : SUCCESS in 41m 49s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"89b3556a63973d51e2cbb7924ea1223a7c8f3e84","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-02 04:53:12.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/458/.","accounts_in_message":[],"_revision_number":12},{"id":"5694df0cbb6961da1be20bd08cb1c795aa9fb392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 05:35:34.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8923e906cb44ed8dca3c4292e62d64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c3906da4d44518a7544561dc7714db : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21754e2b4df6475887000793dd5b1e4d : SUCCESS in 10m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f682e6b29af54e228dd2f7a7a5cc03e1 : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9f2b9e31a804c9da3acc068d1f6ba4f : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d59bf76fd63409bb359a2c9b1a3214f : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/718de23908da455683230b0f06800ab3 : SUCCESS in 1h 03m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89e795653042403482942170ffb1b734 : SUCCESS in 1h 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/313b64fdbb04488c9f6d794f23d18f5c : SUCCESS in 1h 01m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da71548f951a4ba2aac9689da105e86d : SUCCESS in 16m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c57d8593ed2f4a5abc7941b2718660b1 : SUCCESS in 14m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d8546ced56924a3c8ca7cc0bcb570791 : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d29d0e7e313d4b47b36db441441ada21 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7db841a710834c09afabc91bfd2683bc : SUCCESS in 6m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5fd72c73acfc4994baa6c4a3bd9444b2 : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7f96184c8c7c40f7b0a3c1eee3b94e93 : FAILURE in 27m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/689aaa34c5454312973b320506ee1f61 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf55a8a3d5144325a3f07bb461b0de8c : SUCCESS in 1h 23m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/43cf07a98e6b4b1d8461ba8029646ab2 : FAILURE in 2h 14m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1a1a63a5c944af39f995edc2e6aecce : SUCCESS in 1h 27m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e2a5c415b0a84f0c8dee3a9cd2a77ea0 : FAILURE in 16m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2dd614665f764fc397307db20d42b831 : SUCCESS in 24m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7631c02c0e1a43709d5743f583ee3f8b : SUCCESS in 1h 02m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7423243f0de34201b143472549e440f2 : SUCCESS in 1h 26m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9d1292e4fa7441a7abc95df63aef56ab : SUCCESS in 29m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1e1af77cb12042498cbdc8a96ba692e0 : SUCCESS in 1h 31m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/306f50b75ef64214a9fd07a1a6fe4ce3 : FAILURE in 20m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75579b74e87d4b45876da8e5dd768afe : SUCCESS in 1h 12m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f70577080a54d059f8a4c26aed388cd : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":12},{"id":"57312fd171231109df52f85f7f276a5a2feb488b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-02 06:13:35.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c84b6a67d0944b35a49ad5b507b1949f : SUCCESS in 1h 06m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/03fbb5c4f0434b3381aba89bb186f38d : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":12},{"id":"6c5c0992ce4132478c90c4c7b168c5e2fb0e083b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-02 06:40:48.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/913204/12/upstream-check/cinder-tempest-plugin-ontap-nfs/acbcffc/ : SUCCESS in 51m 08s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"06093ca1795acc8f7bd294275330f1feeeecd5a4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-02 06:59:11.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1046/.","accounts_in_message":[],"_revision_number":12},{"id":"62d8cedc06e07bf88e6cb2ef410f89a06cfda7b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-02 08:18:42.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/913204/12/check/DellEMC_Unity_FC/8a5917d/DellEMC_Unity_FC/2526 : FAILURE in 2h 37m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/12/check/DellEMC_Unity_iSCSI/645d2a6/DellEMC_Unity_iSCSI/2531 : SUCCESS in 2h 00m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"864ad518e8fc35980aab216c69249f5d23d6bf73","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-09-02 08:45:50.000000000","message":"Patch Set 12:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/12/check/fujitsu-eternus-dx-iscsi/e64ac68 : SUCCESS in 2h 34m 36s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/12/check/fujitsu-eternus-dx-fc/e3db93c : FAILURE in 3h 52m 34s","accounts_in_message":[],"_revision_number":12},{"id":"c3103e5ffe1b9d425ef75c31b484d8bdf20d7fcf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-02 10:41:40.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/04/913204/12/check/DellEMC_PowerStore_iSCSI/8dd795e/DellEMC_PowerStore_iSCSI/1696 : SUCCESS in 1h 20m 41s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/04/913204/12/check/DellEMC_PowerStore_FC/66dc9f9/DellEMC_PowerStore_FC/1894 : SUCCESS in 1h 04m 56s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/04/913204/12/check/DellEMC_PowerStore_NFS/2ce115f/DellEMC_PowerStore_NFS/1389 : SUCCESS in 1h 07m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/04/913204/12/check/DellEMC_PowerStore_NVMeTCP/450b415/DellEMC_PowerStore_NVMeTCP/1249 : SUCCESS in 1h 49m 58s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/04/913204/12/check/DellEMC_PowerStore_NVMeFC/597da3e/DellEMC_PowerStore_NVMeFC/213 : SUCCESS in 1h 14m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"9e3092457bdddff6b543a58d718f37cce9db2db3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-03 04:08:43.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/12/20240903092649/NEC-V-ISCSI : SUCCESS in 3h 40m 08s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/12/20240903092649/NEC-V-FC : SUCCESS in 3h 41m 50s","accounts_in_message":[],"_revision_number":12},{"id":"7dc206969d69763f4d1f2e9fb6bb1352af62843e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-03 17:13:00.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/12/20240903202159/NEC-ISCSI : SUCCESS in 5h 41m 34s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/12/20240903202159/NEC-FC : SUCCESS in 5h 50m 58s","accounts_in_message":[],"_revision_number":12},{"id":"7de3de78359b18a43a9a893692f8ce773e159ba1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-10 11:55:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/913204-12/HPMSA-iSCSI/ : SUCCESS in 3:17:19\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/913204-12/HPMSA-FC/ : FAILURE in 43:21.88","accounts_in_message":[],"_revision_number":12},{"id":"9bbad4e7c3edd222b03a1f8a7ce280af947870d7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-09-11 17:42:21.000000000","message":"Patch Set 12:\n\nBuild failed.","accounts_in_message":[],"_revision_number":12},{"id":"abe13dcaf2e35c83af53cfa574e96ecf1c808c61","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-09-13 22:19:40.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/913204-12/Seagate-iSCSI/ : SUCCESS in 2:21:07\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/913204-12/Seagate-FC/ : FAILURE in 42:54.81","accounts_in_message":[],"_revision_number":12},{"id":"35cda28bad2779ce751cf0e4d09495c79f555075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-07 06:16:28.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"fa6c91ed6f057a12a62e887b9e2acaf9cf22ec6c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 06:16:40.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1152/","accounts_in_message":[],"_revision_number":13},{"id":"5ea1012fa64aec87c9399a66520ba687cadeae8c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 06:17:30.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913204/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913204/13/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"35ab4aa576f87ea36677323d9372796ce18603d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 06:45:35.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60d9bf6c65d54442a9d9f89344bae922\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/feff33d3834c4c25bd320bbe2bdfc5b7 : FAILURE in 21m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27e248dce01348f9a66d5c97ff167d8b : FAILURE in 26m 47s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8e488a6ec339a4d7dd5c37183d761e5ec0ee524f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-07 07:47:18.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9c352fd92702461cb4ab119c5856ba39\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ee1a4b14ad174dd4bd6568188726351c : SUCCESS in 43m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf42a38c10054087a0e8f2d875496c14 : FAILURE in 39m 17s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a1ead658ff94d6887eb3292c3e5a3b3 : FAILURE in 1h 26m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d19d1baa60e24674be28a11e0f436e9e : SUCCESS in 38m 32s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/27200c7bbaa444f99915a02ce3a51b76 : RETRY_LIMIT in 44s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/55691a06e5a04a19947760c640296194 : NODE_FAILURE Node request 200-0000020992 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb31bf7cb24244afbae5de6067201ddb : RETRY_LIMIT in 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d9c21925a08495f9a3763a6117592bf : RETRY_LIMIT in 44s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devsta... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"7ac6cef414b8e864c8aa5e5673def0833e2b33d7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-07 08:00:34.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913204 : FAILURE in 34m 29s","accounts_in_message":[],"_revision_number":13},{"id":"88d38daf59139fff5e04967cac3b221c3c077b41","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-07 08:09:18.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 34m 54s","accounts_in_message":[],"_revision_number":13},{"id":"2348a24973383387592a2420d99b4f3a398fce96","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-07 08:14:32.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 37m 54s","accounts_in_message":[],"_revision_number":13},{"id":"cfaac7f162ba1c59f614e63d46d6b40cc4713c7d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 08:24:17.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/190/.","accounts_in_message":[],"_revision_number":13},{"id":"28900ed93fad51d142921dad3cdfaafc3a195d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 09:10:59.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa55e64608f84a1d81168489e260bf4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16db9e1130c4298bae6b47e98f5bb8b : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a89f708519d4548aefba349f2d90dcf : TIMED_OUT in 41m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b050951bc2fc44bc830d57e48f9de373 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fab643acec148f49c96379d9e3e5595 : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/0e90152eb62048fba2f1b27520b3548e : SUCCESS in 1h 06m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ee2106691d34bd3ac622b7dd7172852 : SUCCESS in 1h 07m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7a49a2cc185416a8484ffe95143eb64 : SUCCESS in 1h 33m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f42b1315715b4b649a723dc6ac5ad7bd : SUCCESS in 1h 06m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af3099452de64c73a49f8fee025cc45f : SUCCESS in 21m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28951f2dc54e4206b8c0d12f3ff6244c : SUCCESS in 23m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9b8093e4603b480e9bb9dce62adc8082 : SUCCESS in 3m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/674880f0b96c4738b355736172774397 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d1c5c66e2a84480a757a20bcb06087d : SUCCESS in 6m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/826e301c20fb4ef9a8df33644a2700bc : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/abd8654325824a3b808914b29122707d : FAILURE in 35m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6264f0500b5e4ee8b6e9eabfeacc2f97 : SUCCESS in 4m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e7c2d7ff3e2a470f8ba30b437a896ef1 : SUCCESS in 1h 38m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9e076298739d4ae9a5e07992d3dfe035 : SUCCESS in 2h 46m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9da55687639d4edb9a9a19dc46ce1f07 : SUCCESS in 2h 19m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1f88b6ea0b374aae80c8c9844e8c8836 : SUCCESS in 2h 18m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d024e76ed6234211a78ef6963d47cc41 : SUCCESS in 36m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6d7b5e3b20b64c779be2a3ddf5029ea0 : SUCCESS in 1h 16m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6bfc3bed4768403d975ca099c3760c46 : SUCCESS in 1h 39m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/25e42821e7474bb38b580fd6d0eaeb2f : SUCCESS in 46m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34767d3d29854a94a030cc2d2965f444 : SUCCESS in 1h 26m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ff85520447014b93bbec5dedaf3e28a2 : SUCCESS in 1h 30m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce47ca4c72a84c5793782feb254622ce : SUCCESS in 1h 55m 52s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/793ccfa1a0814771a1152d14589c96c5 : SUCCESS in 1h 12m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4a862187a1b4d52921e36a06ed5fcf5 : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":13},{"id":"90ae2848bee4fa577dc9d2cb29b551ddcee83633","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 10:04:10.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/179/.","accounts_in_message":[],"_revision_number":13},{"id":"660c80caf6bfca2fcc5776a8692e45f31d682ed6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-07 10:20:13.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c9aa66a21c6c48c983def9016ce6a81d : SUCCESS in 1h 20m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/111c0ab4a18c433b8ef5e8ab89627cbf : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":13},{"id":"7f90ce966c747dab06b74fd559ee574ac916c95f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-07 17:12:52.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/04/913204/13/ : FAILURE in 1h 19m 55s","accounts_in_message":[],"_revision_number":13},{"id":"5299ace63f2ce8b565440a976afce5d9dbcd0342","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-08 00:22:46.000000000","message":"Patch Set 13:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/13/check/fujitsu-eternus-dx-fc/a52bccb : TIMED_OUT in 3h 34m 17s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/13/check/fujitsu-eternus-dx-iscsi/3619198 : TIMED_OUT in 3h 41m 11s","accounts_in_message":[],"_revision_number":13},{"id":"80d38575b1697d862bcff9b40b5b302bea2b4979","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-08 06:00:37.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"7c182e0b0672304b8e01a84d879d9950d34a9c50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 06:37:15.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72dc207aa5834a40984fd09da921ff92\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a50adcc2b2844675bd4d2c67092f06d0 : FAILURE in 25m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef8adee7003f411ebc9edcdc71819284 : FAILURE in 33m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9065576782b945829ef61c00ee2158472893dc58","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-08 07:27:43.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6a120916a26d498aa73141307d4fce0c : SUCCESS in 1h 13m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1de26232a9834a978b837651a2ec674b : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":13},{"id":"bc1d4f3689c7e3b42dbef1d489939be7f98d3043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 08:22:21.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d69838dd2bed444aac387b9ff2ec2337\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b141378f51d041a59f52d3fc7ad38d10 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f28dc6ce9d2465d9c3fdc74e3fb1cfe : SUCCESS in 9m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/009c22eb619e4109ab3a85856c21089a : FAILURE in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d2e2e874f5345ca92043e798f71c26e : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/21a2f50aac3d4441a1c8044f1e945160 : SUCCESS in 36m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6e385669a904931b581022186581df2 : SUCCESS in 49m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9bc102c3074f4cf8b54f1faa0bcce5da : SUCCESS in 1h 18m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83655fd87ed344328781c723c7fa64b0 : SUCCESS in 39m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b795157a22945f18628d8a5d184d71e : SUCCESS in 13m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0383e6fd4f144b75a803ea640f592329 : SUCCESS in 22m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ff9255fed7f84d4198d94ad8e4a9b8f8 : SUCCESS in 3m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a5febe4e68d4543b94e2fae7f82428c : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/06319dc3703e41d5a4844876d1989abe : SUCCESS in 4m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4c585ecbe186476f9815a0ee34d8a0bc : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ecced7d9d5c545f1beac05e9597bfad7 : FAILURE in 20m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19b43eed21874c9da669b967a0ee0177 : SUCCESS in 7m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d4d88efd63d64faab0ab61bdebcfc8e9 : SUCCESS in 1h 59m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf42338da789410bb72e8d75168837c8 : SUCCESS in 2h 20m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6f4f4af5f26944309455d8cb7f600a03 : SUCCESS in 1h 59m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cce9eb546ac547bca7382d205c8150f0 : SUCCESS in 2h 06m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0e2d446c9b30453792b1a78f50eb77f1 : SUCCESS in 37m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8361b4f34a20473b81f29de970c5a940 : SUCCESS in 1h 14m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/026d3563a9dc4da999ee5fe7fdb35716 : SUCCESS in 1h 37m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/16d38dea49444749ae195701c05db60c : SUCCESS in 53m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8b2f0bec6def48dea036fe90433124ce : SUCCESS in 1h 17m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/890a74449a5d4cbdad86ba052cddf2e8 : SUCCESS in 1h 59m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8ab788ad39e24f60856c3f01c4228431 : SUCCESS in 1h 51m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00d78318cf514724ae836e218c5d7fda : FAILURE in 1h 38m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8338bfae9340463f95745566b5aac66a : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":13},{"id":"31017e4eb31a00a536b5e5ee461e832968b402e0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-10 09:41:55.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/13/20250110142837/NEC-V-ISCSI : SUCCESS in 4h 06m 05s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/13/20250110142837/NEC-V-FC : SUCCESS in 4h 13m 15s","accounts_in_message":[],"_revision_number":13},{"id":"70b991f741aea3541098f3abf66469bdaf444bcd","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-11 09:17:54.000000000","message":"Patch Set 13:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/13/20250111100136/NEC-FC : SUCCESS in 5h 48m 05s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/13/20250111100136/NEC-ISCSI : FAILURE in 8h 16m 15s","accounts_in_message":[],"_revision_number":13},{"id":"e8b77ce140e9a19d2e1a2099d74728cc82a5fc13","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-14 08:35:06.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"800ae3c0fd898b496d39b43f55e86e0532b03b48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 08:55:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dedc84237d7c4eca98f553380e94c14e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6f764d5b59f4ea28fe121fb23884cda : SUCCESS in 17m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eee81b0a25094f848829a9c7938b0086 : FAILURE in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9fbfe0924a5bf706072e69fbd2b5f757128c54e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 11:10:57.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d43a6620121340adb72a07f66d4dd032\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0f7b47983542bc905c4af2f946c54f : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87114d6d19334522ad785c9572dbee4c : FAILURE in 16m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99d1a9ce05ce43898c5ff24c690266b2 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ef1146fc264ae2b033d0e24323b7b0 : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/9caf0f8c2cc2480a82ebd64f3b24813b : SUCCESS in 34m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c80de3e5917e4a57a25b1783808fd359 : SUCCESS in 1h 07m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6eb3821f519246edb49b4502fa7590fe : SUCCESS in 1h 28m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed82ccdb062a499499f7666a2ecd077f : SUCCESS in 1h 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6685759c8f64ab5819e8331fc1cbe0a : SUCCESS in 17m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c36f27ef69b64a449b1b6d1e7fa0160f : FAILURE in 21m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6671a5076cc749a2aca188155ae832e8 : SUCCESS in 7m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a2b8b80521540e1a72f3383cea8bed5 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/80fec44457dd4aa0a3db75e3169de2b7 : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cacaf66f151e4790b29fe8bb14c59628 : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f7a98a0144347b2a3e89bb2cb3a57e4 : FAILURE in 34m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f212530576264235a8aca6ea57bbb52c : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1dd0e510879545d7b28074c58aa6dfbe : SUCCESS in 1h 44m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0b89fa3374d8497aae481106f34e45fb : FAILURE in 2h 29m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a5d510fec54a4a3ea4fdd00426b2635c : SUCCESS in 1h 29m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/39d9ac21e2f749c882f56f38a01ed543 : FAILURE in 2h 09m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6cc5440f85aa4db4a9e469e39f4f2016 : SUCCESS in 36m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ea47637b43b34b35910a5f770adfcb3e : SUCCESS in 1h 00m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1065d657d084cc28157572351f33483 : SUCCESS in 1h 43m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76f0a4e8b888496eba4ffe7de81e31d9 : SUCCESS in 51m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9bdf7a1f5827487a9c2c006ea199bb24 : SUCCESS in 1h 55m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e81959de482241999db900d8c5ef127a : SUCCESS in 1h 34m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89345b4f79f647fea7c9fbbfd1635dd7 : SUCCESS in 1h 41m 36s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0b829e3dadb4f12a2fbcb40b2b6f59a : SUCCESS in 1h 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d418d5a14724436387df2bd2f30001eb : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":13},{"id":"cef200468f98c55ddca92efaf9225447245510a1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-14 11:17:01.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9bb97d105d664fdda757101ce5a15ce4 : SUCCESS in 1h 23m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2d2c336ea0d946db8912cf6c279f05b4 : SUCCESS in 1h 33m 38s","accounts_in_message":[],"_revision_number":13},{"id":"1f8ddae071ecfb2c0bc15321edbe5ee50a429acf","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-15 02:17:25.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"52147e62ffc65bb0ab88dc455c8636e0462bc3e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 02:58:12.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c9917c6e77249d793501d2888eb39c7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1500ccddc6641e39c62c716eb2435a3 : FAILURE in 33m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e408bfb078b45b2a09352fa71bcc13c : FAILURE in 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8bc4a907bf3504b488ed313ba9041776a53ebeb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 04:51:58.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee3de281816d47f8901adc98b0f07b2c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01abd1a7ff97493fbfbe9c4e356bada0 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d755421df2b4cde9e6134bd2c2e2a22 : SUCCESS in 10m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d62a646061844f880257803d7338074 : SUCCESS in 14m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b200563d78db4a1daac6cf35716853d4 : SUCCESS in 12m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/562021d0e487459aab789a25ee8cb126 : SUCCESS in 54m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3451401fb7df4b77be7c04a5075d6b8c : SUCCESS in 42m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/20b79770cfc34e649f2d7c7939533c13 : SUCCESS in 1h 29m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17f9b18a84514dfa8d63ec76d8542e25 : SUCCESS in 54m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b178352701745dca6df95eccbfa96e2 : SUCCESS in 16m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/08676d9d6ecd456a92459ecb6498d3a0 : TIMED_OUT in 41m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a2ffd328947459490233c6e113a4231 : SUCCESS in 6m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1943bcba2d5c40ea9486839ec52bf2ba : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ddcb8beeac4487398bb61192d120d80 : SUCCESS in 6m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1c3396276d5f46a892efe0a1f3de5057 : SUCCESS in 9m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6805fb662cf948d9930a05e525058b34 : FAILURE in 19m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a70e40f35122426b99299020b8de35dd : SUCCESS in 7m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3020ab74e8fe4e37b5ba359e82c125aa : SUCCESS in 2h 08m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a05d5b56c5a441b386c66960bdff61cb : FAILURE in 2h 24m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5b4a831a86a14bd5a1a26cd7852bf4c5 : SUCCESS in 1h 36m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/15afb3fbbe20489f861a37272557904e : SUCCESS in 2h 06m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9ebff395df814d0ba9888360229ada42 : SUCCESS in 38m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d6eb4b0ad92e495e979a760543adf0ed : SUCCESS in 1h 12m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ea8f6fd421b547298e6a57e6bf55cd81 : SUCCESS in 1h 25m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1518484354544735991a1fb1e18537a3 : SUCCESS in 56m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a020b73396bc42eda87789bbbeeae154 : SUCCESS in 1h 40m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0babda402db04d0cb86ae54c5b23ece8 : SUCCESS in 1h 34m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe058a52a94a4c44841c6de2b6d02803 : SUCCESS in 1h 59m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5008f93c9784f75964c92953059f2ec : SUCCESS in 1h 37m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8137a1199ce845779c6603ed7878c1da : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":13},{"id":"0884a990a54fb19309b46a4b81780be63f72938d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-15 04:52:10.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b820a404be254457b7c9f824fed3fcce : FAILURE in 1h 17m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/80431e634c2f4282b44bf7ef8a44b35a : SUCCESS in 1h 18m 56s","accounts_in_message":[],"_revision_number":13},{"id":"2771ff20d3c125536eb4f78b9368fc0e721f9ffe","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-01-15 04:52:55.000000000","message":"Patch Set 13:\n\nBuild failed.","accounts_in_message":[],"_revision_number":13},{"id":"f0c2033460f6dbec11e5bdef4f19485efc4849eb","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-01-15 04:53:12.000000000","message":"Patch Set 13:\n\nBuild failed.","accounts_in_message":[],"_revision_number":13},{"id":"568a6402d6f4df426ffb3320171c37e238ed33c5","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-01-15 04:53:39.000000000","message":"Patch Set 13:\n\nBuild failed.","accounts_in_message":[],"_revision_number":13},{"id":"ce34bb4a38ca896f1d902454d328034d4f042907","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-01-15 04:53:39.000000000","message":"Patch Set 13:\n\nBuild failed.","accounts_in_message":[],"_revision_number":13},{"id":"aa2d38e8f571e7805038c20b36929db08986ab09","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-15 04:59:07.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_01_2025_04_53_14\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":13},{"id":"16c8b617a1132ab4bc3a46632423485d676f5119","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-27 14:05:55.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"f004b1eb2ed2f3a50824a10555b8aec5db6eb3b7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-02-27 20:00:08.000000000","message":"Patch Set 13:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/13/fc : SUCCESS in 1h 19m 39s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/13/iscsi : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":13},{"id":"512543a5f0e09227fa1e6868985e98f39e24be0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-21 07:37:52.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"e02cccca3d4aaa17cd5602114af31713c1454a75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:38:01.000000000","message":"Patch Set 14:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1546/","accounts_in_message":[],"_revision_number":14},{"id":"27b2478036b6e5855d2ea36bcb9339a095d39cf0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:38:05.000000000","message":"Patch Set 14:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913204/14/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913204/14/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":14},{"id":"102d2f023689b564dee690b450cec587dea197f0","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-21 07:39:50.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"42dbbd36b86fa47555aee9659d2990650925e9cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 07:53:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ddd02110cb74e4f836a4f084ebf5463\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2bab6626c0e4765bf57e2211d8cdebb : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7cfa27195a048a48b011b541ba41174 : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"909dc9ed6e4da3cdff97903d093920366f5ac62f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:21:20.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22aef751e750473b89b4122a48b873b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1778cc0c3c44c3b9cfa781da972aa04 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/247cb8db557c4358937169bf6897117a : SUCCESS in 13m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5d34091c16b45a1884663918d616d62 : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4015111d136472bab7dbb8ff8a0b372 : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/6be50227c3f749459fb3a9e3afcbe8a1 : SUCCESS in 57m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/31de23f8a6eb4fd18359651b085f7eb0 : SUCCESS in 45m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e376c6b69374b8d9a32f22f1915a3ed : SUCCESS in 1h 30m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f6a11db92f4457a8e369b8f7017156d : SUCCESS in 59m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cd7d14d20f5450bbfb8c4ad22d3e673 : SUCCESS in 21m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/63f70f0d10f347d195ed5935e660a093 : SUCCESS in 16m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6859e53ba4b4d1aa96bfb8d6317c15a : SUCCESS in 7m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5ffc230024a4110824821fc981c7467 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e399933c487447038bbda626f195dddf : SUCCESS in 7m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/826f31daf3384b2f8b58f4498ace2cc6 : SUCCESS in 7m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/465335821ea74307822c13225e7e4f77 : FAILURE in 34m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49383303f14d449ea171cfa4dec3aa88 : SUCCESS in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6645648641694d22af1689980ffc4dbd : SUCCESS in 2h 03m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c366c43274cf41478cc6b009a3b011f6 : SUCCESS in 2h 37m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/650ff45d8e4b4fc6b31e238c8597c3da : SUCCESS in 2h 21m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b5c7de6d25d7417cb0dc71d81c1bbd29 : FAILURE in 12m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4233e174d10f4c0f9a6b7b0425d9b080 : SUCCESS in 35m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34c99c81b6f843bf85c17ce8496fe1f0 : SUCCESS in 1h 12m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/05b4c9b6110c42cbad75ac8ad0536858 : SUCCESS in 1h 29m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/485d49ae761d440498531e2db0e5b0d4 : SUCCESS in 36m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/23f4b190dfbb45faae27c4a1d14ad1df : SUCCESS in 1h 27m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0e00a793e7d5418d9681af65808d25a9 : SUCCESS in 1h 40m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb1d4dc6127242a58bfa9fb6282b3bb6 : SUCCESS in 1h 52m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a1cb79344ea4064a7bb1e8ff90994c8 : FAILURE in 1h 24m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf964d5b55ff45e39d48ab3311acf6f1 : SUCCESS in 49m 26s","accounts_in_message":[],"_revision_number":14},{"id":"0599e3f4d67e9cd6bd8c6c9959be39a89f1b05ff","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-21 11:23:30.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4f7578ba8f2d44979aedeb574520d50d\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8438199d6d4c4f289fb3d06415cde41c : SUCCESS in 45m 46s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/55335317e5c949d1a48c7b7dc4f08a34 : NODE_FAILURE Node request 200-0000033751 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e3e49950ab634711ac2de6e802f09728 : FAILURE in 1h 44m 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0f0c041b1f274c41bfca4e25617fe8a1 : SUCCESS in 44m 28s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4b7721e5d7934936a7bacc1a88a4a137 : RETRY_LIMIT in 44s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8479d8afaf0349ae81a2c315b1af17e9 : NODE_FAILURE Node request 200-0000033787 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/31c61f8f80e94563ac698e6c8ff36732 : RETRY_LIMIT in 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/69b7daaa96b0448cbf3bf0fb874bf693 : RETRY_LIMIT in 37s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n           ... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"fe5ae22278c4d9ca69c0f52571d04c078549d8f8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-21 13:29:24.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 33m 49s","accounts_in_message":[],"_revision_number":14},{"id":"a1ec3dd2be7f08d65ac0e9dc61f3ceb3f427c1a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-21 13:36:17.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 34m 03s","accounts_in_message":[],"_revision_number":14},{"id":"24231e89dec408fe15b8d0c331045739e0611922","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 15:28:15.000000000","message":"Patch Set 14:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/552/.","accounts_in_message":[],"_revision_number":14},{"id":"88dc914e2e61bcee4e680100548c48c32b2b2efd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-21 17:11:46.000000000","message":"Patch Set 14:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-14/FC/ : FAILURE in 8h 25m 23s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-913204-14/iSCSI/ : UNSTABLE in 2h 27m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":14},{"id":"b73e4be57c9eb753741e6f46b34377c74a333d7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-21 21:59:18.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/232669e6059144e69ef2457ffdd8e8e7 : SUCCESS in 1h 09m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7fe1fa8baaeb401398daf5b108255649 : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":14},{"id":"a31dd6a76c9f81ebc9150e0bd97b4bcaf79e46fc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-21 22:02:28.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"7ba1f16941f73d88217ad799e6c012d53ac2ca0f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-03-22 13:53:31.000000000","message":"Patch Set 14:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/04/913204/14/check/fujitsu-eternus-dx-fc/bdab239 : TIMED_OUT in 3h 34m 40s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/04/913204/14/check/fujitsu-eternus-dx-iscsi/fe9f796 : FAILURE in 2h 15m 23s","accounts_in_message":[],"_revision_number":14},{"id":"daf12c94610ba6baf9cbc0482e681f4378412c89","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-24 05:25:19.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f5ce05d94208b9d5e35844301ea06e9f80994f28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 05:40:58.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b18894cc3bee41fca35e40a45848e3dd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f25dcf4c4c1044b39df2f9bf10d063f5 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec78b0de29974419bb8b39257aabb2f4 : SUCCESS in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f9b04fc3acfde8509ef0714ab4c86cc8d36c8950","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-24 06:46:19.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cc3507fd10894549938d204c9e7fe0b7 : SUCCESS in 1h 13m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ba36023488264aefa9cfb6b18f5dac93 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":14},{"id":"1633068cd600df0ed168b6540cb4e944921e2773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 07:51:46.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b35a32ea6d7148d58191655445c9e27a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c747c3a601c4dc6ab9c3eaa634444a7 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/190f7cee17804a02b4f9e083a0a55c25 : SUCCESS in 7m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bc0153b5ecc4f0297e6a7db509e201e : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe513309c4204a018f2760f82d7df168 : SUCCESS in 6m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/bb7c0eac93c146c7bc39cc0c96c691b2 : SUCCESS in 47m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/456e77828221494f9497ed73d25028e4 : SUCCESS in 25m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/efb60656f3b3416ba91dc7f2c20ec141 : SUCCESS in 1h 08m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4579b6433a5a42f9bd8058b13d72e9a3 : SUCCESS in 37m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db8cbee7f1374ed5b8f932b496cad651 : SUCCESS in 15m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/170642fb824943468240b230471e045e : SUCCESS in 20m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ddca8547c854448a8c5600595b42656 : SUCCESS in 6m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/33f4c129fb6d4d8097dfbd757ea414df : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5078449005bd4ddda76391da55c19a3b : SUCCESS in 3m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f4b15ae5fc5484b8352f8684b70619d : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d9a7574d6c1246d4bead1f8cd992cea6 : FAILURE in 37m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/24ee2b3b6a8049cbb0f8feba988744ed : SUCCESS in 7m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a3ed68e44a4a4221b31ffb292bec1ae3 : FAILURE in 1h 53m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1ab05c9006f7464e9de1d861b677b802 : SUCCESS in 2h 21m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2210c13ac3d54647aa3fac98ffc5da1f : SUCCESS in 1h 46m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dc99174e6bbf403ab93ac0b3e89e1086 : SUCCESS in 1h 44m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a7dd41988c21436bac4a535cb6edfd27 : SUCCESS in 26m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c1668a4f3da84558b0f271c3a0f72078 : SUCCESS in 1h 02m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1698561408ba46f9891ee5a7cb657591 : SUCCESS in 1h 38m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86f0a5a9133a4c3787c08e3d6ea9830b : SUCCESS in 30m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a9ebbb0b20534e6cb05e8c4c9f4d3611 : SUCCESS in 1h 11m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/49b85c75698442d09b3bebf559e48042 : SUCCESS in 1h 53m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/33a89e5929c14e6ba68fde8ade74305f : SUCCESS in 1h 38m 45s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73e9577c16754800be6500c59477299e : SUCCESS in 1h 16m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acd651173b8946e7801daf1ebdfe0cbc : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":14},{"id":"c1af720ff24d96a75a4d0c42512565aa40a1de60","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-03-24 11:00:07.000000000","message":"Patch Set 14:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/14/fc : SUCCESS in 0h 57m 04s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/04/913204/14/iscsi : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":14},{"id":"f714e836e5ca30cad12dae30b0968c6c92754ab3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-24 20:32:58.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/14/20250325012106/NEC-V-ISCSI : SUCCESS in 3h 49m 26s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/14/20250325012106/NEC-V-FC : SUCCESS in 4h 11m 49s","accounts_in_message":[],"_revision_number":14},{"id":"f114284a511af7165b95b0c4ad979ef05a43b58e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-26 04:03:53.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"3b90390f2ae1c1b258945446b4651968dc3b2143","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 04:04:04.000000000","message":"Patch Set 15:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1577/","accounts_in_message":[],"_revision_number":15},{"id":"2382d127949ca7267ae876a909cc60c974a8a4c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 04:04:08.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913204/15/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913204/15/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":15},{"id":"c7ae13f90070f3604af576195b29be197fdf9e55","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 04:05:46.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"ebc74509ca33863f9620a2f08a1503b8ed5c375c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 04:19:32.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/934075f24dec4a61851159f3210df55c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d50b1fcc5f54d64b076e8b02740fcca : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/233a955710624211a6fae8f8793e7a96 : SUCCESS in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"28e593e5413d2e0e1729694f7d224851f7f2478d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 06:40:04.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90427f40e4a2490b8196809ae59b266d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f673e0801949ef855ce33ab4cf9fce : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e389012d1ad34a3cb83855d416f28c9d : SUCCESS in 8m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b3acf79850c4915bf218c03a06283ce : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a739f1c40804503bfb203aa811c97f1 : SUCCESS in 7m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/3bb9aabe5ee8472ab57fda3d53e6ac58 : SUCCESS in 46m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57aa373d3f6740dc9594cdd52d393106 : SUCCESS in 46m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d14526906404f94bf2bd341f942e859 : FAILURE in 1h 28m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8d783900a1a4ec6adea0e04072e3d42 : SUCCESS in 37m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40d9c5da31094e4286649942a0aa6c2f : SUCCESS in 17m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/89b1e661bc5e4715a038024c679fcec2 : FAILURE in 23m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6764eebef046423ca30e39309522d248 : SUCCESS in 7m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7438ba7f4f440fb82ae385433df681e : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d2fbd0312864fce9c555c076b4caf10 : SUCCESS in 6m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c55d389ff8e04f63905c20bf296a6eab : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4d71b38f99b04d13bef7309e1e5dbc20 : FAILURE in 31m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49b4b3bfb890424c9bbcf2d734b845d7 : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f914671151cd4df487962bd0a4c56b0f : SUCCESS in 1h 52m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/03c9e00ffaf14aa698dcf712322464a3 : SUCCESS in 2h 34m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cd854ff5189b4df289fbb30670d9bb1a : SUCCESS in 1h 53m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/eed844006921480590132e41f48e78a5 : SUCCESS in 2h 01m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bdda7e8fc77242da84fbdfc25d0eee88 : SUCCESS in 34m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/87aac78cdf0d48a5804cc8c67d929770 : SUCCESS in 1h 10m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6006a176fa704af28ec6c473073d2467 : SUCCESS in 1h 26m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c9dceb5b11f413a8490f568b2bb8b3f : SUCCESS in 46m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a935ef62dfe142cab4ac41c8f837d100 : SUCCESS in 1h 40m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6205b4451bd2470da50341e71874f00a : SUCCESS in 1h 49m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb6729e080064dd3a8ebe7d38a1836d0 : SUCCESS in 1h 34m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17a819059df445b4985715e89d63c88a : SUCCESS in 1h 20m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15d510baeefb40a28cd97c010a4a796c : SUCCESS in 30m 28s","accounts_in_message":[],"_revision_number":15},{"id":"e13164c8e2c865e37130b2367b1c3b191f620f8b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-26 08:45:20.000000000","message":"Patch Set 15:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cba3b83e6baa4dd9b055606e377466db\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a9d2a7dc435b4671bca802ee78bbb127 : SUCCESS in 42m 27s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de87c0685b054668bf35b8a6308265e0 : NODE_FAILURE Node request 200-0000035656 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/87ccf6a2618a4587a6cd3a412988a158 : FAILURE in 1h 56m 23s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/63b09288069442acad3e73c33cde8753 : SUCCESS in 38m 09s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3268d44ac8334866a49aeaa92d9e5ba3 : RETRY_LIMIT in 41s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43d883205650495999145a66efb2e592 : NODE_FAILURE Node request 200-0000035660 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b34fe48faf974272a4461703a7519519 : RETRY_LIMIT in 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90396a44b57d4dd59647a6971f5889ed : RETRY_LIMIT in 43s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.ya... (truncated)","accounts_in_message":[],"_revision_number":15},{"id":"b29a768ba26dc794519e3cb2efde9433c221c3f7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 11:05:48.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913204 : FAILURE in 33m 08s","accounts_in_message":[],"_revision_number":15},{"id":"e1e9af33ba5eb4b9981ec8002e95a2290ed01bd8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 11:23:59.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913204 : FAILURE in 33m 49s","accounts_in_message":[],"_revision_number":15},{"id":"60b44a4b5f1b1b26fbd9d37fd91fbad6602d461e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-26 12:29:43.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"4f231046dfc6eed7777c71eb704f50662af1b3cf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-26 19:06:12.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/580/.","accounts_in_message":[],"_revision_number":15},{"id":"555b8f085365b2d1b4d8765b730072b1edf514eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-26 19:44:06.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e397f0d3e0cc4828a8f010b69941bf54 : SUCCESS in 1h 20m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ae5da5501faa4e95bb78f551dd33c2bb : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":15},{"id":"133da95c833f594b3a9fe8ffa5a3d885dc4eddcc","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-31 08:10:45.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/15/20250331125822/NEC-V-ISCSI : SUCCESS in 4h 11m 50s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/15/20250331125822/NEC-V-FC : SUCCESS in 4h 12m 20s","accounts_in_message":[],"_revision_number":15},{"id":"7da09ee4d08847873e937bb4fb1b19a17d34c1d5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-01 06:22:37.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/15/20250401092851/NEC-ISCSI : SUCCESS in 5h 29m 59s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/913204/15/20250401092851/NEC-FC : SUCCESS in 5h 53m 43s","accounts_in_message":[],"_revision_number":15},{"id":"f3427ff229b2f0bf835a8dcf7070026f055ac7a3","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-04-03 23:58:24.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5b4041768cf4da18f5b20c61c7c5c1d9b4a5def7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 00:19:05.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01c57cc4b9be4f0997ab9cf2a22709c7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4e068d2cbda4cf2a886ebd869b7916b : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28f50a89a6e4465d8d92f7d05bf3f55d : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b823feebbc4242f993301c49fc55e46c : SUCCESS in 18m 14s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"37633e4f346b8f4e5bb66d5138dbed6efde59cd1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-04 01:26:19.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fa1d99e0435e42d2953373968c78e6c8 : SUCCESS in 1h 24m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2f7205a0113944f994f1ac8a3eb9b91a : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":15},{"id":"3be17d4f8865aef749631166fe0a058323e412a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 02:33:31.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f13d4628b4b0472a9eb1ccddfbdf5ce1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2319537e3587433b9ffa40bba74bca86 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f5caa9780654bf4b91f62f0b1cf5c86 : SUCCESS in 8m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f670e05a0b304b02878966ed346cc5e7 : SUCCESS in 13m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e87c7c7a34374a849445c2eac503a332 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0224a8125cd420d819c2d16108270e5 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/2554568af83d4b9498a0ed18bc5a8dc5 : SUCCESS in 1h 01m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a2ba8dbd18b542efaca68ed7c14e96bc : SUCCESS in 47m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc3244af5a4c47b5990e7d3728df2062 : SUCCESS in 1h 31m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a155bd2690e4728899fca93d132fe1b : SUCCESS in 48m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/affbedd7674f4bda8f690bbeb705bc87 : SUCCESS in 14m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/37ab6ce06a004558bffd28cb7af241cf : SUCCESS in 22m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b708c60683ee46b995187231892bb93a : SUCCESS in 6m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1f1142e1b614b0daea73d18b2f55751 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9337fe5104974621a810ff552840527c : SUCCESS in 6m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/32a0489982fb444c838681ea4a105543 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc8b6f6deda54003bf19eb751850777f : FAILURE in 36m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a22ea7140fe74dd48ffd2ec237695eae : SUCCESS in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8fccf42122f44d6791de130e281bfbd2 : SUCCESS in 1h 34m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0eda7d4f0f154722a51de22d1280d4f7 : SUCCESS in 2h 32m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/620169424cf3493daa4b60fd6be1b357 : SUCCESS in 1h 46m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/43e301ebcf9b41259246541780164f58 : SUCCESS in 1h 31m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5f9360dccd974184b6a34391051e0acc : SUCCESS in 34m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5637cea0f02e4120b3aa396d05391992 : SUCCESS in 50m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08767515460c4715be4d31072bdd98fa : SUCCESS in 1h 25m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aa6b4c5dd2fc437a9a460a621075e04c : SUCCESS in 40m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/35f86e0104ca4bafb9742cada599a5bd : SUCCESS in 1h 14m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3384f2e15e67454693723e1e296a3d5f : SUCCESS in 59m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/613443cadcb947d7a2397a34580ff557 : SUCCESS in 1h 25m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/495c0e8ec11c400bb00d87d5fb723a7e : SUCCESS in 1h 14m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e676fb2b9104efeae736231ffa24a1a : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":15},{"id":"269856483f9b3c22ac9c2dfb405977867e5d7250","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-04 05:41:38.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"01be1af98f32c84a3b90a5816a3aead9d94f1288","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-04-09 05:25:45.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/913204-15/PVME-iSCSI/ : SUCCESS in 1:57:11\n- PVME-FC https://os-logs.seagate.com/dell-ci/913204-15/PVME-FC/ : SUCCESS in 2:07:04","accounts_in_message":[],"_revision_number":15},{"id":"5e74121b6ac5f8d53af7809d78c46969f15bf05f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-09 09:50:05.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/913204-15/Lenovo-iSCSI/ : SUCCESS in 1:56:40\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/913204-15/Lenovo-FC/ : SUCCESS in 2:06:27","accounts_in_message":[],"_revision_number":15},{"id":"f5357dd5a78441052407caeae29f409de3537842","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-04-09 14:14:23.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/913204-15/Seagate-iSCSI/ : SUCCESS in 1:56:27\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/913204-15/Seagate-FC/ : SUCCESS in 2:06:38","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"0d2a6321e0bb466211051f5e360f7b317131838b","revisions":{"cb236292756e5ccdbde9e14f8dda43454fb54502":{"kind":"REWORK","_number":1,"created":"2024-03-15 06:09:06.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/1"}}},"commit":{"parents":[{"commit":"3b68f194bc5625e5a56a59e4187fe5a53a322fa2","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b68f194bc5625e5a56a59e4187fe5a53a322fa2"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb236292756e5ccdbde9e14f8dda43454fb54502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb236292756e5ccdbde9e14f8dda43454fb54502"}]},"branch":"refs/heads/master"},"1329fa860e598cbdd1c013c7bd68655bab6adbe1":{"kind":"NO_CHANGE","_number":2,"created":"2024-03-19 06:27:27.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/2"}}},"commit":{"parents":[{"commit":"5554585c44a19065362570786fdfe70010edf878","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5554585c44a19065362570786fdfe70010edf878"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-19 06:28:03.000000000","tz":-240},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1329fa860e598cbdd1c013c7bd68655bab6adbe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1329fa860e598cbdd1c013c7bd68655bab6adbe1"}]},"branch":"refs/heads/master"},"4d83884f1b5c30ca26a4d03cba0c517e3b4f531f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-19 09:42:50.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/3"}}},"commit":{"parents":[{"commit":"0ad473d0fc94b3d31a3ac50dccbebd8e188f86e1","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ad473d0fc94b3d31a3ac50dccbebd8e188f86e1"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-19 09:41:45.000000000","tz":-240},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d83884f1b5c30ca26a4d03cba0c517e3b4f531f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d83884f1b5c30ca26a4d03cba0c517e3b4f531f"}]},"branch":"refs/heads/master"},"7d162c4b3aeee9a226679527ed9f2050ea17e883":{"kind":"REWORK","_number":4,"created":"2024-04-09 10:14:09.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/4"}}},"commit":{"parents":[{"commit":"820ccc7e277143fdc34b94f3ffbfe44c570ca12e","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/820ccc7e277143fdc34b94f3ffbfe44c570ca12e"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-04-09 10:13:34.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d162c4b3aeee9a226679527ed9f2050ea17e883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d162c4b3aeee9a226679527ed9f2050ea17e883"}]},"branch":"refs/heads/master"},"0ea7d896010781927626ee15a3dfd65f53f849cc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-21 05:31:27.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/5"}}},"commit":{"parents":[{"commit":"676a71c4134bdd5cbe3a720b9643a90d69f276ff","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/676a71c4134bdd5cbe3a720b9643a90d69f276ff"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-21 05:27:39.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ea7d896010781927626ee15a3dfd65f53f849cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ea7d896010781927626ee15a3dfd65f53f849cc"}]},"branch":"refs/heads/master"},"28b4787527c10976ccca71cd19c63ef816cd028a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-24 10:13:56.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/6"}}},"commit":{"parents":[{"commit":"b8e3cf55fd92bf237f20e54ad6c99cfb66add32f","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8e3cf55fd92bf237f20e54ad6c99cfb66add32f"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-24 10:08:23.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28b4787527c10976ccca71cd19c63ef816cd028a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28b4787527c10976ccca71cd19c63ef816cd028a"}]},"branch":"refs/heads/master"},"88b174193217de7d2e269e1f16d432a985833094":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-05-27 09:55:39.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/7"}}},"commit":{"parents":[{"commit":"f4c2fb088dac47ec2a57e6ca545bbe2ab4be6be8","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4c2fb088dac47ec2a57e6ca545bbe2ab4be6be8"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-27 09:55:13.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88b174193217de7d2e269e1f16d432a985833094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88b174193217de7d2e269e1f16d432a985833094"}]},"branch":"refs/heads/master"},"f18af5162f71e0d4e0880ccd1602964eb6381fc2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-06-14 07:01:43.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/8"}}},"commit":{"parents":[{"commit":"0ff7f708fe7ea21c51848b857404107b81dd7fbd","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ff7f708fe7ea21c51848b857404107b81dd7fbd"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-06-14 07:00:19.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f18af5162f71e0d4e0880ccd1602964eb6381fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f18af5162f71e0d4e0880ccd1602964eb6381fc2"}]},"branch":"refs/heads/master"},"ce302e4e5ed50bd73833b02c2087ea2a5fbc32d8":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-07-12 02:26:09.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/9"}}},"commit":{"parents":[{"commit":"b98b340d08dee84527ca48e74c4ac47a315ce0e7","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b98b340d08dee84527ca48e74c4ac47a315ce0e7"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-07-12 02:23:45.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce302e4e5ed50bd73833b02c2087ea2a5fbc32d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce302e4e5ed50bd73833b02c2087ea2a5fbc32d8"}]},"branch":"refs/heads/master"},"57dda2e1c2c7114fde7a07d7983ede752b5a2bec":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-07-12 09:47:58.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/10"}}},"commit":{"parents":[{"commit":"0e8f9a1fef653bbd38dd3d39da49e3f0484aed2d","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e8f9a1fef653bbd38dd3d39da49e3f0484aed2d"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-07-12 09:47:34.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57dda2e1c2c7114fde7a07d7983ede752b5a2bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57dda2e1c2c7114fde7a07d7983ede752b5a2bec"}]},"branch":"refs/heads/master"},"9eddb1dbe6ef252aaf49e6601d6471ff5f331ad7":{"kind":"REWORK","_number":11,"created":"2024-08-06 06:48:17.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/11"}}},"commit":{"parents":[{"commit":"de38750c1e3b8ee661f26815dd16ec6380b1ffbb","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de38750c1e3b8ee661f26815dd16ec6380b1ffbb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-08-06 06:47:31.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9eddb1dbe6ef252aaf49e6601d6471ff5f331ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9eddb1dbe6ef252aaf49e6601d6471ff5f331ad7"}]},"branch":"refs/heads/master"},"1e605cb8d16e74ca9651e520b910ff6b4a26a623":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-09-02 03:15:18.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/12"}}},"commit":{"parents":[{"commit":"f2119764caeb79ef9aefbfac51a17df179029b4c","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2119764caeb79ef9aefbfac51a17df179029b4c"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-09-02 03:06:16.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e605cb8d16e74ca9651e520b910ff6b4a26a623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e605cb8d16e74ca9651e520b910ff6b4a26a623"}]},"branch":"refs/heads/master"},"c8a0c8ecdfe57053e6c33013a50ef4b6456e6bb0":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-01-07 06:16:28.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/04/913204/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/13"}}},"commit":{"parents":[{"commit":"c925f98503dca8ba093a003c42aeb5fa430f2e4b","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c925f98503dca8ba093a003c42aeb5fa430f2e4b"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2025-01-07 06:15:55.000000000","tz":480},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8a0c8ecdfe57053e6c33013a50ef4b6456e6bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8a0c8ecdfe57053e6c33013a50ef4b6456e6bb0"}]},"branch":"refs/heads/master"},"1b540895112ec1418bb206ed1ceefc1d56c36aa7":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-03-21 07:37:52.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/04/913204/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/14"}}},"commit":{"parents":[{"commit":"6e1f1cdb388ea7bb0dc63a8d1eb6dec522670cbb","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e1f1cdb388ea7bb0dc63a8d1eb6dec522670cbb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-21 07:36:45.000000000","tz":-240},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b540895112ec1418bb206ed1ceefc1d56c36aa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b540895112ec1418bb206ed1ceefc1d56c36aa7"}]},"branch":"refs/heads/master"},"0d2a6321e0bb466211051f5e360f7b317131838b":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-03-26 04:03:53.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/04/913204/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/913204/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/913204/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/913204/15"}}},"commit":{"parents":[{"commit":"1fd22f4abc6999d876b43e3ec8aa202dc014b41b","subject":"Fujitsu Driver: Added support for timeout.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1fd22f4abc6999d876b43e3ec8aa202dc014b41b"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-15 06:09:04.000000000","tz":-240},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-26 04:03:37.000000000","tz":-240},"subject":"Fujitsu Driver: Add support for retype volume","message":"Fujitsu Driver: Add support for retype volume\n\nAdding support to retype a volume in Fujitsu driver.\nThis patch implements:\n- retype_volume()\n\nImplements: blueprint fujitsu-retype-support\nChange-Id: I8fdbc8087115761d796f6f5fd85253df47a35049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d2a6321e0bb466211051f5e360f7b317131838b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d2a6321e0bb466211051f5e360f7b317131838b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
