)]}'
{"id":"openstack%2Fcinder~913217","triplet_id":"openstack%2Fcinder~master~I9168dd609a42b0f60dd210f1184fb50065f409ed","project":"openstack/cinder","branch":"master","topic":"fujitsu-driver-update","attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2025-11-12 03:43:25.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2025-04-11 10:27:34.000000000","reason":"Someone else replied on the change"},"37834":{"account":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"last_update":"2025-04-11 10:27:34.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9168dd609a42b0f60dd210f1184fb50065f409ed","subject":"Fujitsu Driver: Support manage volume and snapshot","status":"NEW","created":"2024-03-26 01:42:47.000000000","updated":"2026-02-02 08:44:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0e205687f17b33eb5a30057ee3ccbc497f93a0b8","_number":913217,"virtual_id_number":913217,"owner":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-11 07:10:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-26 01:57:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-26 01:59:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-26 02:07:32.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-26 02:33:03.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-26 02:36:56.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-26 04:23:00.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-26 04:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 05:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-26 09:33:36.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-26 11:20:49.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-26 16:41:43.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 14:09:04.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-19 15:45:48.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-19 16:48:07.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 06:32:59.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-14 17:22:01.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-06-15 12:01:58.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-07 07:43:45.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-04 05:21:27.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-03-24 05:27:32.000000000","updated_by":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"reviewer":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"state":"CC"},{"updated":"2025-03-26 15:54:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-11 10:27:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-04-11 13:00:57.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-04-11 17:26:51.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-04-11 21:55:20.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-10-31 18:02:39.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"CC"},{"updated":"2025-11-12 03:43:25.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-02-02 08:44:17.000000000","updated_by":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"reviewer":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"state":"CC"}],"messages":[{"id":"586647ec404bdb4125670a46cebe4d7e2ccdd66f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-26 01:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3a91d0f2862922653b418997cc0b34da420b9cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-26 01:57:30.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9efb8733e8c249498743d7a0b7559fdb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6b0f3f9af6674324856c440362af7199 : NODE_FAILURE Node request 200-0000002880 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"e362d9424ad503f56d32b2d8dba88fcb411d5daf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 01:59:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b8b5036ddd495daa57d7b71e2d91a5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37fe451a6d6a4496910c6c76c16f2937 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2de07eb00c54e25be272b7ad4a72b05 : FAILURE in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74310ceaefd07b35c0a1e1ba384233bb56c41277","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-26 02:07:32.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-cinder-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"98d8d4960c0a46fd3514ee6380b78f9e34b6b2f9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-26 02:33:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 45m 04s","accounts_in_message":[],"_revision_number":1},{"id":"37b216437a0413a85070891177f7bdf9eda25983","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-26 02:36:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 48m 50s","accounts_in_message":[],"_revision_number":1},{"id":"bdd82c5f9463277f92a56cc32bcf76b317fadcf7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-26 04:23:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/17/913217/1/check/fujitsu-eternus-dx-iscsi/b41aba5 : FAILURE in 2h 21m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/17/913217/1/check/fujitsu-eternus-dx-fc/18027c9 : FAILURE in 2h 32m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6a60165d5535f70aa697a19857a6e9b12d737384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 04:24:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca064035601c453690a4bd78d8edbc8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d57bc5f98504f778828f951fc925a45 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/528693a60cf3423586bcab5d16aab264 : TIMED_OUT in 46m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfa5ffb399d644168cc66b1281607861 : TIMED_OUT in 43m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1370c849a4c14abfbcbfe221a3e61481 : FAILURE in 10m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f699f1945b4d5c8b7f187e8a0ba2b5 : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/75dd0cf7daf542ee9775f5dd0729dbff : SUCCESS in 49m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a78cfd5c3e4d4994b697a3dd9ef6061c : SUCCESS in 48m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce567fcf4d1d4a4b9bae37352eabf4f9 : SUCCESS in 1h 29m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93b1c4c8fb364792bf3e7118ab1c24b4 : SUCCESS in 50m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/107f06519fb2413d912894f73a3c576c : SUCCESS in 14m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/375120787c894ac78bac551471a080fe : FAILURE in 22m 24s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/554ede9078d04f2e84a909c90295bd21 : FAILURE in 39m 36s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ddbdbf89e20745a7bdf240871ec74121 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f46d063af0f248bf8eb8a2b7866be193 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/97dafa751e4f44f29bc0a74f5944d123 : SUCCESS in 6m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01faed23810549c88353877516c84394 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/84bf7548d80d4ca5a613eb6fbf2e7b95 : SUCCESS in 32m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b4c991e57954ed4bb0bb5786c92e6ad : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/929b86d96be54279bbe1ef125abbf65d : SUCCESS in 1h 33m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6175a2da110644f883fb386361831d41 : FAILURE in 2h 30m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f252a78f324e4541bc8b911d070ff6b4 : SUCCESS in 1h 37m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/60f86eb47635402ab058a3a8b6fda7ff : SUCCESS in 1h 41m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9b4dfff90742413d8e176bad0185dbd6 : SUCCESS in 28m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/22a020a12a4a46b98d02f1a2d410d8fe : SUCCESS in 1h 00m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8329059e2534cbca410597a28914362 : FAILURE in 2h 04m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a2570697a5d440ecaf31c4978ed15e7c : SUCCESS in 46m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9da3d2c30aa24527b4b32fe8a697bda1 : SUCCESS in 1h 19m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e275209295304091a2cb3ed8c664ccc8 : SUCCESS in 1h 46m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2149cb203acf4724835dcab9f7d5da12 : SUCCESS in 1h 49m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4486b334d62439c992314aadaa18a00 : SUCCESS in 43m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c15ef4a33d874f6a6c876e482dbf2a9657e9c9ef","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-26 05:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/17/913217/1/fc : SUCCESS in 0h 47m 03s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/17/913217/1/iscsi : SUCCESS in 0h 55m 08s","accounts_in_message":[],"_revision_number":1},{"id":"cedd3f2864374405b2d77aedf964c91535386268","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-26 09:33:36.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-1/FC/ : FAILURE in 7h 50m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-1/iSCSI/ : SUCCESS in 2h 19m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"01bccc034659d921d3924a2f2bce2b3eb2433970","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-26 11:20:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/1/20240326164940/NEC-V-ISCSI : SUCCESS in 3h 15m 09s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/1/20240326164940/NEC-V-FC : SUCCESS in 3h 31m 05s","accounts_in_message":[],"_revision_number":1},{"id":"803b745a9cbbbb9d90b3584093426a4f10ba8d93","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-26 16:41:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/1/20240326170120/NEC-ISCSI : SUCCESS in 4h 11m 37s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/1/20240326170120/NEC-FC : SUCCESS in 8h 40m 20s","accounts_in_message":[],"_revision_number":1},{"id":"171db137c18bbb684145da659c8e7cae82f92181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-04-19 09:37:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dbf78baaf844e45c09f4a01e80d7180f741d1b94","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-19 11:49:54.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/78d52d4f863f474dbf667c896c017580\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e7ecee382f6d4f9997b8d6356c3e399f : FAILURE in 1h 38m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7075763d75134e24918de0f66fc25b9ee77ae423","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-19 14:09:04.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/913217/2/check/DellEMC_Unity_FC/1a7fbd0/DellEMC_Unity_FC/1699 : FAILURE in 1h 32m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/913217/2/check/DellEMC_Unity_iSCSI/545ee56/DellEMC_Unity_iSCSI/1730 : SUCCESS in 1h 35m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8d79bd6c853c24b83ef19f7ae5dbdcc8bec6c774","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-19 14:15:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 1h 26m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2a7cbb812710cca8e5ff6df75fa20806b06914fa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-19 15:45:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/240/.","accounts_in_message":[],"_revision_number":2},{"id":"cf3b2275bbd908af5211cf391f29c462900599af","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-19 16:48:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/17/913217/2/ : SUCCESS in 1h 13m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0f07f0c645ebb9654f22b7b3c494b603e6745919","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-19 19:21:01.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-2/FC/ : FAILURE in 8h 59m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-2/iSCSI/ : FAILURE in 5h 31m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"18f9b3994aabf10acb07dc807ea3462e0e987c2f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-04-21 14:50:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/2/20240421201333/NEC-V-FC : SUCCESS in 3h 23m 01s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/2/20240421201333/NEC-V-ISCSI : SUCCESS in 3h 37m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5840413a076d7f4c92182b7a4d428583fe122b96","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-21 23:43:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/2/20240422001054/NEC-FC : SUCCESS in 4h 47m 25s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/2/20240422001054/NEC-ISCSI : SUCCESS in 8h 32m 51s","accounts_in_message":[],"_revision_number":2},{"id":"13abf19a4bd7d00b9ba64f6e5d22873009ea02db","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-25 06:32:59.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"dc4c4f313e79c4e2d9afb49746683878becf4e23","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 06:33:56.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"1a4b491bdbfe9e2262b851a235700a6b97c5581d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-21 06:14:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f61864dcef2891a566f32525964b3403b84c4dd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 06:31:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31997f1353414e17a25da32209edb4cc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9a502a4158c4e4e8d1de66700e32632 : FAILURE in 14m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/913bed51302947a4b4c18c75ae3d5244 : SUCCESS in 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45199ae925e2f94c58ee74aac842bf4aaf081737","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-21 08:46:47.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 25m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ed43eca5fa1f14d3467f64bd8f73796a2d02bb3b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 08:51:05.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7fd4687ba1ba4d298343d2c77b52b956\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d1330534d9ac495198f97cb0328d29cb : POST_FAILURE in 24m 34s","accounts_in_message":[],"_revision_number":3},{"id":"0f1b207795fa110a6ea34cc18d73e9990b46fbb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:51:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0996d2c3f20e4f94b853acd20f8afde4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee43bbd2a1c642d28bc45dd5abafc079 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09f4a8429d5e4c179cbae6d41a1b301a : FAILURE in 27m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d5ff5b3c11e4983bab501f310e33ed8 : FAILURE in 10m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fc7e7e8c15b4f1eab9292f85be534dd : FAILURE in 10m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/027ccb666df840cda64da793bc684c37 : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/6143121d02ca4a039cca4ca9aaa8980c : SUCCESS in 52m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc4105a134714d35b5e5b7e8e756b9bb : SUCCESS in 1h 29m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8fd148a8e78437b9f5bdf31d640b13e : SUCCESS in 47m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f150f16c71b34474b5216d5ea427577f : SUCCESS in 12m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c183475d77ed43fab32ec9985cc98ed9 : FAILURE in 26m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2f5d7674c14740209d7a3f2db0caa79d : SUCCESS in 6m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e831517a04854a968b94acd7be8832c0 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fca8acac4b154421a8f338218922de37 : SUCCESS in 6m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/034185695acd4497bc7fb1a3955a97f8 : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eab689193186491b91b9f38915964fb7 : FAILURE in 51m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dec23b40338248c7a288de51780934d8 : FAILURE in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8012f3b45c654f298b649e0656c1b8c2 : SUCCESS in 2h 01m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cc05bdcc03844a94afae718ae98fece3 : SUCCESS in 2h 00m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/35a7ec291a8049adbe1c99779ca59307 : SUCCESS in 2h 06m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/061dbcde92214249b11de9cdba827927 : SUCCESS in 2h 04m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/53c954bc0340481da32adf5c0837de66 : SUCCESS in 36m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/71df65f4f22c4c2fb24b92adfbf52d83 : SUCCESS in 1h 06m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97b1d593ed234bdbbea495701df57a48 : SUCCESS in 1h 33m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fe2e572234eb4f4d92b81433819f93a3 : SUCCESS in 48m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/30f930cb51b04d409cf3b050dda8a093 : SUCCESS in 1h 25m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b4dbde51564949a3a7a09410ed6aa299 : POST_FAILURE in 2h 31m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b3b9132176f43dcb46914859db93c39 : SUCCESS in 1h 44m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f6e3e17340a4e87b783a0eb3f638d0e : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ff9a19546c1f4323686c292f9af61e75968cf61a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-21 11:22:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 35m 55s","accounts_in_message":[],"_revision_number":3},{"id":"0947907ac3664b7b52d4e89e853177413b4b8bb5","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-22 05:46:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"538d553887002af844d1e17f0695483e22cf897e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 06:16:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2ac6fec17e54859b1fd339d8c375a79\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fde3c5f1ac1b42bdbbc6ac0a1951c220 : FAILURE in 26m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/84ebdb3962c442dba641a89de701688d : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dfc0b1040470b8ed7cbc8821fee8c568d4d6da0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-22 08:32:38.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3ea48c18bddd44a999303b859d4f8500\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a56d227a7f6d4ae684525d938c928b78 : POST_FAILURE in 24m 23s","accounts_in_message":[],"_revision_number":3},{"id":"3a22dd78ff9b3ccbb3f955088976ca37884b3f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 08:32:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dc5ddfbb8fd4b1fa89c315f45d82ba5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/779fceb91514423a8fc88f0ce384114e : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c4a3ac0f7a2434ab174154d866b3a8f : TIMED_OUT in 1h 13m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1497aa6a6f924852905a8dfe4db7096b : FAILURE in 17m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e25b2881f6b94e0ea70302e4bff99877 : TIMED_OUT in 41m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/206da69163624212acfdb6da5fe07dc9 : SUCCESS in 13m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/101b87c870b34f8287fa882d79d05286 : SUCCESS in 1h 00m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7050a2bd0154573b704ae2238392114 : SUCCESS in 1h 32m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10f3632d51794640a50b5df838db7e42 : SUCCESS in 46m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/020bd12ec1864697a4996d035034305b : SUCCESS in 13m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f9d5c11da394368824fcf1300a818b1 : TIMED_OUT in 43m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/98ed5380e2be4e318f5e556e26fd6492 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e41bc25edac44fe1b436f67a20646574 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/efccd60f0a5749c99e72ee3dd0937e3e : SUCCESS in 7m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/671e71b2305a4dfbb38148ed514a1ec6 : SUCCESS in 7m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/36d3aaf35e4a42cfb5d3e8f46ef4fa4c : FAILURE in 45m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a70666b6f8841caa2d6abf572dc2776 : FAILURE in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc239826abaf4bb8a5db1fe2eae03734 : SUCCESS in 1h 56m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/89beb7315bad43c0823cf5828565b43b : FAILURE in 2h 23m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b6b42361ccda4097b53d243e1a346ca8 : SUCCESS in 1h 55m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d3f663e33a1c4236a1f825e090360183 : SUCCESS in 1h 44m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bee21b06ff634fda9ac35e7f52fd1e4f : SUCCESS in 36m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c07a3a30959e489c927b4bc98045844b : SUCCESS in 54m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d996157601534f20bbc913e3cfd3a7df : SUCCESS in 1h 31m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0b1221789b714bf1b1471595ca440cc6 : SUCCESS in 50m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1249fa08023d47e789566d78352662d8 : SUCCESS in 1h 27m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8edc5a89393643648a679921434a4659 : SUCCESS in 1h 32m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1713090b554c42a2a1474412f137ed7f : SUCCESS in 1h 59m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d832a7598f254d1690bb393499046de1 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":3},{"id":"3370629eb4ce7bed694e17ffa64f19f8e8396d38","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-22 11:02:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/456/.","accounts_in_message":[],"_revision_number":3},{"id":"3f0aaae6fab0ee7b2cc456a80a655ea5b99332b5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-23 18:46:11.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/913217/3/check/DellEMC_Unity_FC/fb7bc85/DellEMC_Unity_FC/1936 : FAILURE in 1h 55m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/913217/3/check/DellEMC_Unity_iSCSI/ec959b8/DellEMC_Unity_iSCSI/1964 : FAILURE in 1h 56m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"7bb0f75cbe6c631687e906a9bce4b1d26144a809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-24 10:22:52.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c09b61ff4d0c0b3f62f9b00af55753ac342f6967","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 10:39:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2e72ece657d451590c2e2477433454e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0128b441065841ffa6b17d04cbff6a0e : FAILURE in 14m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a93bd0082e0e4001ad1b58b2822e8e9b : FAILURE in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6cbd48a06a20eb99905e958ac2a6a6338d70984f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:32:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1a67f8cd987437c9fb6a22e8f628893\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5c329c6bc54e1ca08121266b010beb : FAILURE in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7f1528fec4d429ebd81d755837f96b4 : TIMED_OUT in 1h 22m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0bf5b1e7f1a46b08c54513b88b4fa90 : TIMED_OUT in 1h 00m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3e90860563b4de591f3746472d4c099 : FAILURE in 20m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909f951d85144ed5acff4071f32df273 : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/15531e968f8a4136852c4ecb5d176d5b : SUCCESS in 49m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b185d245fc01405f89c8899880678480 : SUCCESS in 1h 40m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/914dfc1000474152ba162910c282fb82 : SUCCESS in 56m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2eb565e5b784b549a571dc649f57911 : SUCCESS in 13m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5e2dd765e85d4877b24fc0327ae3c835 : TIMED_OUT in 48m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/69ba3a9687a14c33a88f23e0ffd00bd7 : SUCCESS in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32b0eacde1194061bc665caf0af1e9df : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2075c8a6e29a4e4aadae09404e875fdf : SUCCESS in 6m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/70ef971dd4e44fd28e42e08808812e94 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5f0ee65acc11498db7456ff9367696c5 : FAILURE in 34m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab90446f86074850b7d74a3725ae77ce : FAILURE in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70b075a3556e488eae9728b1a3372ade : SUCCESS in 1h 45m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cf211671686a4d87a516fea22dc4c3c9 : FAILURE in 2h 06m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2c8c9d8c6f134371a90096adb5f85460 : SUCCESS in 1h 27m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4f6cb10bae0349319116c92bd02a43ec : SUCCESS in 1h 57m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dfce10628d3c4e9e9bfaf712338f0aba : SUCCESS in 37m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ac29da8dd004247acf2a9d6b74ad0d2 : SUCCESS in 1h 15m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/60928a77cf574464914dd8f4878308a1 : SUCCESS in 1h 22m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f21c736ee2364b96a652940b3e78d58a : SUCCESS in 44m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b3d2a160a0954bf78bc454d093124474 : SUCCESS in 1h 22m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0aa3a9abd8124c7a91c6e573b53ff7b3 : SUCCESS in 1h 27m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8bf74e68d483471dae7a13870c7b376b : SUCCESS in 1h 49m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75368c20e1204a34ad213e2f39bbd325 : SUCCESS in 40m 55s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/fujitsu/eternus_dx/eternus_dx_cli.py","accounts_in_message":[],"_revision_number":4},{"id":"1344b4118a0acd59b07a26305f27bcd192ef608d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-24 13:40:53.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/bd9594123fad46b797abfbbaf76c6196\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c73ba23667f145bab02222bdd63335ed : POST_FAILURE in 30m 43s","accounts_in_message":[],"_revision_number":4},{"id":"0cc493dab572d357b957dbc4fc922b725e4d0478","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-24 16:15:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 22m 02s","accounts_in_message":[],"_revision_number":4},{"id":"8287bb8880788bcbe79fe7741dd1556523d514a2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-24 18:55:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 34m 33s","accounts_in_message":[],"_revision_number":4},{"id":"816df9131cbcc6ee4a97cc430d09845b78a83a4e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-25 08:50:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/498/.","accounts_in_message":[],"_revision_number":4},{"id":"b27c6d7a967f5a1f6cc92bfdfb472fe3f9e79ef4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-26 07:06:19.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/913217/4/check/DellEMC_Unity_FC/957315d/DellEMC_Unity_FC/1971 : FAILURE in 1h 25m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/913217/4/check/DellEMC_Unity_iSCSI/2ff7bad/DellEMC_Unity_iSCSI/1996 : FAILURE in 1h 44m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"678ee056ebf3a5c17b42ad6aca6fbf75eb849497","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-26 07:28:58.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/17/913217/4/check/DellEMC_PowerFlex_3.6/0c2759a/DellEMC_PowerFlex_3.6/1452 : SUCCESS in 2h 00m 01s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/17/913217/4/check/DellEMC_PowerFlex_4.5/32c6cf7/DellEMC_PowerFlex_4.5/366 : SUCCESS in 2h 08m 29s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/17/913217/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/8bcb1d9/DellEMC_PowerFlex_4.x_NVMeTCP/306 : FAILURE in 2h 55m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"626b1875b124d091a12ee04277730e51b143d78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-27 10:04:57.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4eb888fc5275c9bfe52676b29d4f3cd5bb5a50c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 10:28:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df4e620f9fc84f9991395050973dc7a5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcce45229a1149b4a11d1fde59988d21 : FAILURE in 13m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9dde2cd8e3ba4fcc925cc165fe95025e : FAILURE in 20m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2bd3bb6cecdc0da78439e987357894fd5dafef6d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-27 12:15:53.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/38b88d50b5104e518428375e511e4a38\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/67d292cbd0d9472297e017b6f4500ead : POST_FAILURE in 21m 52s","accounts_in_message":[],"_revision_number":5},{"id":"3d9b4e35fa5d7233f4e3559299684496dcb6135b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 12:36:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97a4b2b625c849faadadf8d752647562\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/816fcdf61db349d9befbbdcfbcda7ac2 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7020b43678c54346ab1d774d6cb8df62 : FAILURE in 19m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/530894b5c8da4887af6fedd079b16391 : FAILURE in 12m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c11e17335fbc425d832c83e989759c92 : TIMED_OUT in 1h 23m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e215df3658ed46219ba997f08e683c66 : SUCCESS in 13m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/2a3208ab7f754863a2e0920927f4b70d : SUCCESS in 1h 06m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f3d95b017ce46d3b70687474f2dc113 : SUCCESS in 1h 28m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e358a1a1ccad49ea97faf332beeb7505 : SUCCESS in 59m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cbcc520e174498a8a55e3d72fb873e9 : SUCCESS in 12m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a9138bda974e400499a6d27340f8ad11 : FAILURE in 37m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f29674c9ddc4b708fd1c9d5cd08b018 : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/73f34c041db645d0a3e50c57122d3da4 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c14ceef3b5e4c4ca26150df0eacb951 : SUCCESS in 6m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6297e59a536e489196b6e6ea9d8f6d9c : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9273b69200c84f76a19de4e88b411273 : FAILURE in 40m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1baa071ea9d84bf7b6b1cca231f4f108 : FAILURE in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2240f21443e34737aba0bf5060520426 : SUCCESS in 2h 12m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7f96666c49494fcf890c62021288d287 : SUCCESS in 2h 24m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f2f2cab1cea2492c947bbfdf5537fd99 : FAILURE in 2h 04m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b2063b2d4dd14ff9a8fe4b3a649bbba2 : SUCCESS in 1h 58m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4626e448e9af47148942269dc92ebc3f : SUCCESS in 37m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce2b2d4db0f74b94bcb684c9500ca99f : SUCCESS in 1h 04m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ca4db0734fd4470ab657207aaf71d531 : SUCCESS in 1h 27m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0d169bd4f7b14ef1a47987cfe5806fa9 : SUCCESS in 30m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2afaf08e7b7d496d91d8fe8896e7801a : SUCCESS in 1h 37m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b117c872b2944bc5acb4645ebb6b8394 : SUCCESS in 1h 32m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54b505f634314db686bdf0f138d903e8 : SUCCESS in 1h 30m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af50deee8aac43f69f5a1919af2b17e7 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":5},{"id":"35c07275a5f01a376d3d3a6ba53fa35ddb28d10f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-27 19:00:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 36m 00s","accounts_in_message":[],"_revision_number":5},{"id":"6a37a8d6c15e2792e294353e38fa1c2691e94d2f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-28 04:15:28.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/519/.","accounts_in_message":[],"_revision_number":5},{"id":"878ac13816d882ea6d1c4a9656380d6ff6074ae9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-28 16:36:58.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/913217/5/check/DellEMC_Unity_FC/ffe4f46/DellEMC_Unity_FC/1992 : FAILURE in 2h 10m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/913217/5/check/DellEMC_Unity_iSCSI/288cb14/DellEMC_Unity_iSCSI/2017 : SUCCESS in 2h 21m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"21c51b112eccb0a36b320dd4eadeae634969506c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-29 02:07:53.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/17/913217/5/check/DellEMC_PowerFlex_3.6/706c91f/DellEMC_PowerFlex_3.6/1473 : SUCCESS in 2h 04m 38s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/17/913217/5/check/DellEMC_PowerFlex_4.5/c203269/DellEMC_PowerFlex_4.5/387 : FAILURE in 2h 14m 10s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/17/913217/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/daeb9be/DellEMC_PowerFlex_4.x_NVMeTCP/327 : FAILURE in 2h 55m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"1c78a25f770ca8e5dee51ec7c1815d58bdd0f011","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-06-08 15:47:44.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/5/20240608162538/NEC-V-ISCSI : FAILURE in 7h 34m 02s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/5/20240608162538/NEC-V-FC : FAILURE in 8h 22m 03s","accounts_in_message":[],"_revision_number":5},{"id":"2a0a8e6d5bbe2b9a4239fa1410d662c02c80ffb9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-06-08 18:06:18.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/5/20240608172942/NEC-ISCSI : FAILURE in 9h 02m 42s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/5/20240608172942/NEC-FC : FAILURE in 9h 36m 33s","accounts_in_message":[],"_revision_number":5},{"id":"ecccdf4650265872da2043cfb1f071971e14b512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-14 07:23:24.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e9fd3061a7cd8c85d07beb22e88f8bac60f1eb50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-14 07:26:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"1df2b0ebc295bc57cbe122d49ea4b93a935404e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 07:43:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bd685a1efb4433fb9d955a53c87ff37\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2396e5de859d41868ddc7d88d06b830a : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7f16fcf176647e59bc7c9317cbdb628 : FAILURE in 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fd7e15f7b9074a5a6fc5f55c8861d510adbea077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 10:13:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d250bcbd8d3c4154b8a2e313c238cee2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc52962218584b428e1c9ae6958f248f : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91df247111f8477bad3a4b4ce58774e4 : FAILURE in 22m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc4e14b8763f426084242eb15b2d4537 : FAILURE in 22m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/569246bb369145aa944326b9afb490a5 : TIMED_OUT in 57m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b2899caaa81478a8973983c33a87b91 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/dd64671fc955466da8fe9e5c1da877f2 : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2db15271a708405dbcc2b3c27cc8d962 : SUCCESS in 1h 23m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2f27fb4abc245d4a19aafa1bcb4e72a : SUCCESS in 1h 16m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7ee6a6046b445c7adba76579929c558 : SUCCESS in 18m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c71207a0cf3341be916f2af00e668e03 : TIMED_OUT in 41m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bb88c7bda48147d78055f4738028aa27 : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93c7ae094f194422b3bad62e69ec0b6a : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27bc4ffb57ac4c31bf2af8f4ef777bf6 : SUCCESS in 6m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9c036b5b550347c5b709b9743e83f96b : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a041504ec3149fe9759d557dd1746b4 : FAILURE in 40m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8769a3fb3304aaa894255f19a8e3dbb : FAILURE in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9fe970b66b35441f975df4e9dca79560 : SUCCESS in 2h 16m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/052847d5967b4731824e6afbbfde6282 : FAILURE in 2h 46m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5a97bcc0d054779a43259afb7eea53b : SUCCESS in 1h 49m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4fb1b2ef78414ca790c6aa8cb150906c : SUCCESS in 1h 50m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/36c378cfc6c24ed6ac949b79af34f046 : SUCCESS in 38m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8d48cd0f305f429cb89bcec7d8a68694 : SUCCESS in 1h 05m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4be670e119eb425ba96d9811b3c1b719 : SUCCESS in 1h 50m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/37c4933495d547ab80edb0f331bbb470 : SUCCESS in 44m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8ae80f90aeca46898339d7ff691f2544 : SUCCESS in 1h 26m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8020732bd2554abcacc35c1e9b36bc22 : SUCCESS in 1h 58m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f607d05de904c808ba9566f419fbdc9 : SUCCESS in 1h 51m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a1d7490422e4fb7b216e09e4525d231 : SUCCESS in 1h 18m 26s","accounts_in_message":[],"_revision_number":6},{"id":"594de6130d8697d56726dea6906462ede55bb5cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-14 10:21:01.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/78180e596de04d7d9a2b3bef35fe1e14\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/966062ba66a94a27bea8e7c22f6ec96f : POST_FAILURE in 28m 25s","accounts_in_message":[],"_revision_number":6},{"id":"48b514946d9f065c0bfef147b95d6e2e3beb036e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-14 17:22:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913217 : FAILURE in 34m 24s","accounts_in_message":[],"_revision_number":6},{"id":"a4453bb002545fcb1249d9742fbe8f517ec05d62","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-14 18:57:04.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 34m 26s","accounts_in_message":[],"_revision_number":6},{"id":"19860de051e9b321de4096749a7bef9ee5c0cc31","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-15 08:51:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/17/913217/6/check/fujitsu-eternus-dx-iscsi/010b38e : SUCCESS in 2h 33m 31s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/17/913217/6/check/fujitsu-eternus-dx-fc/a3f1e27 : FAILURE in 2h 33m 48s","accounts_in_message":[],"_revision_number":6},{"id":"84705b9574e658f065c42311b1cdd4192f730ea2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-15 09:37:49.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-6/FC/ : FAILURE in 15h 8m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-6/iSCSI/ : FAILURE in 10h 36m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"72c5970eceb17649d344154083ae2719d06f22db","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 11:19:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/122/.","accounts_in_message":[],"_revision_number":6},{"id":"eb02a06e6b625e1ba0bd55b48e56ab541a63bdfa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-15 12:01:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9828dd3e3a97479b81e0b7f32b24a937 : SUCCESS in 1h 18m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4976231ec52f462dbf6212be68b0ecc3 : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":6},{"id":"01fccddaeb50e6492530d0645c42520a0b20eb60","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 13:22:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/608/.","accounts_in_message":[],"_revision_number":6},{"id":"fc585d10ed58f57f0aa85419770b81e1cab4c03a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-16 00:23:43.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/17/913217/6/check/DellEMC_PowerFlex_3.6/1ef2eda/DellEMC_PowerFlex_3.6/1583 : SUCCESS in 1h 46m 03s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/17/913217/6/check/DellEMC_PowerFlex_4.5/2497ba9/DellEMC_PowerFlex_4.5/478 : SUCCESS in 1h 56m 28s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/17/913217/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/708923b/DellEMC_PowerFlex_4.x_NVMeTCP/428 : FAILURE in 11m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"d5605c5f843b6667d3405aa9b3c057ff3a96e63d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-17 13:19:00.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/913217/6/check/DellEMC_Unity_FC/ac1ea1b/DellEMC_Unity_FC/2092 : SUCCESS in 1h 27m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/913217/6/check/DellEMC_Unity_iSCSI/7b86c17/DellEMC_Unity_iSCSI/2115 : ABORTED in 6h 00m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"6f2e0e86a2c234319c783954209e41921da57415","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-21 02:32:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7bee5d3106f484d925a65d1191b17e81cca21c0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 03:05:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf45913aa6c7428b953ba900a513b834\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91d73272f7cf463f9d6920ad00c252de : FAILURE in 16m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/85d0022c7f0e46f3bc97b669206ce26f : FAILURE in 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"da95976b6ffe4b91e504eae388f6e09b21cc3e7c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-21 05:10:33.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fe554e34b6a04f8e88f40d86a00b25e2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/349a12f7f1a94a60882a7331fccab164 : POST_FAILURE in 24m 05s","accounts_in_message":[],"_revision_number":6},{"id":"b6c92bfd4018ecb3b47851965438730ea4bd5cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 05:18:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/334d402ba1a247069fc6d68de35ce2b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df1f2aca9e544cdfa1cf8226163c166c : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74f91916483d46dbad163702a5390437 : FAILURE in 26m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2621fc9936b4731b077fa5c88d9413b : TIMED_OUT in 51m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67099ea982654ae5b5a2998c53822002 : FAILURE in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/881e19a7ad044d9785cd68c943387dcc : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/064a4013739f490397a2f3f3adcd0351 : SUCCESS in 1h 04m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d3e6c734ac3494794c09cfd6a7e7a46 : SUCCESS in 1h 24m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/597ea201fccb45ffb390c17526d37b98 : SUCCESS in 1h 07m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bad28430dc714684b7a6c2f44400a303 : SUCCESS in 19m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59d336c7b9cf41b580c68baea81745c7 : FAILURE in 24m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2bab922a42d34899b549dbf448e4d777 : SUCCESS in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b1b1987154044f6abc54c7a21193cc9e : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ca101204daeb4f35bb9f31113af38647 : SUCCESS in 6m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a02ab85b8a54415ba97aa0c45a9691aa : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/18ddb0fa190c41a0b8f27dc34ebdb308 : FAILURE in 45m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01b87162ec314f3ebb31ffd4f1fd2bd9 : FAILURE in 6m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dc16022604ae487f901014a93f426720 : SUCCESS in 2h 17m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/faebf8165e22435da430d956f56f1aa2 : FAILURE in 2h 10m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f9348ce48b824e21b8768ad4e43a258c : SUCCESS in 2h 03m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bad665c5ec6b491dabf0c2c10f79935b : SUCCESS in 1h 25m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a03debbab1334794bc401889f2147722 : SUCCESS in 29m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aa23ddeec82e40fd8068cd49e15b6ceb : SUCCESS in 1h 18m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5bd443866b44ade828c143b581e9a4e : SUCCESS in 1h 22m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d807cb9932dd490d823e0d8d07ada74a : SUCCESS in 44m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/69a58a9e689742f381cdedf8b7dd3be1 : SUCCESS in 1h 34m 25s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/575c3bc2ef074b02923ac75861542697 : SUCCESS in 1h 39m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc98d4e028aa4091aad3bc7d7b45f70d : SUCCESS in 1h 51m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/109f17bc7d4b4c738eac49a629ddc58a : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":6},{"id":"85791cfe81dc7c53073e503d25f3dcec3be3ba1b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-21 08:14:57.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/157532a85caf463da2fd230b9073ea5a : FAILURE in 1h 17m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/17f2fafa36474fce8b84d76fbba6ca61 : RETRY_LIMIT in 8m 33s","accounts_in_message":[],"_revision_number":6},{"id":"e6a231b873c91f5149de675863e035cbefeb8799","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-24 17:55:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8f7d96acafc64072ba738f2e7713070c : SUCCESS in 1h 21m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/43d72b0fe4d44c7fa268735df4e22db6 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":6},{"id":"1dc1ff3957a44c27412fed8302ae7ebb806d257e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-12 07:29:15.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c279cb587f08795513e84411d349d0292a6446d5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-12 08:00:40.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 12m 15s","accounts_in_message":[],"_revision_number":7},{"id":"424517442d36726dcdac1c4864542a90acffcae0","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-12 08:00:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913217 : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":7},{"id":"e89c6b7711515255ba524e517d384cc5c4d4e451","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 08:15:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ccaeb49120643aba961673478c9fc52\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd3e9b9d721048ae94f6c88e3a5848d3 : FAILURE in 3m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a00e52cd113f42f7b376457464dcd7b1 : FAILURE in 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"76505f5280403bd7e9d1099a48064f6d94d46438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-12 09:51:27.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.","accounts_in_message":[],"_revision_number":8},{"id":"22e11a4609e188808bd12c41f942ea40a11f1246","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 10:36:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec3f5d7d2f742b5b9dd1fd3a1b051c0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2d31adb0386455db9e0b9f0eaecdf99 : FAILURE in 42m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1f4a84750d6e473589b82649f35ab9b9 : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"37e4536a65ae48581999edba23a17a72163b3fb1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-12 11:09:40.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-7/FC/ : SUCCESS in 2h 37m 34s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-7/iSCSI/ : SUCCESS in 2h 56m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"7f6afd0a60c86ad7c5ca13e0c2b3a96b035d14fc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-12 11:17:44.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4ece6420f6144c2492906a17ee5d92bc\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9a411122355142339ff79e7f017376a9 : POST_FAILURE in 23m 13s","accounts_in_message":[],"_revision_number":8},{"id":"c36f25e204aed055a81342c1963cba3978add695","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-12 14:00:07.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/17/913217/8/fc : SUCCESS in 0h 48m 25s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/17/913217/8/iscsi : SUCCESS in 0h 54m 33s","accounts_in_message":[],"_revision_number":8},{"id":"bb01f77c57459ed4589cccefc24cde5bb06fb16c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-12 14:44:45.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-8/FC/ : SUCCESS in 2h 57m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-8/iSCSI/ : SUCCESS in 2h 14m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"f8151da8a91da64e15b594357b0e3d7b7cfdd75e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 14:51:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e92cdab9c5544d5694267fbf3ed45bf6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951fa3801fcc4be199a031105b34e5b2 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb5b95e4aa5e40599d2736cbfb7217a1 : TIMED_OUT in 1h 28m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aca8159d003745c0a3a81594d8745d29 : FAILURE in 22m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa27cdd6760448a0b5777fd314156648 : FAILURE in 23m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9856e53845d46fe911a94ca497c0274 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/079529f06bd94b18ab2767470e614f20 : SUCCESS in 1h 05m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56a69be777614ec5ab4cc65c7b7caa26 : SUCCESS in 1h 27m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/117e810a201947c1988efcf6635ddbf1 : SUCCESS in 1h 04m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3aaa81821a443a2abf7bdf6232f2289 : SUCCESS in 18m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1550ed5d3b2b4515af42f2c214c1223f : FAILURE in 31m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/770cb191350d44a0b6cb58b789050070 : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20de1a2b634047d6806e07203fec98b8 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8870756a7afc4e9f83b053faf73c02c4 : SUCCESS in 6m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/806fab0361054fe1a8d88d9fbf3a7010 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a741aed58a7b4bb693f98a1934dcbb69 : FAILURE in 36m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9c74cf090af649fda63cb82560a2c6b7 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/337eacb96dea4f7aa86d05be6e733b84 : SUCCESS in 1h 48m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ed7174ec3e4c4ffd9cb41f719e08aeb9 : SUCCESS in 2h 31m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9dd1bf581798469689ecd1f36dd9ae5e : SUCCESS in 1h 45m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/636b334662d243129027eb185d4cebfa : SUCCESS in 1h 51m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eed89ce6c70246799163e110ae952efb : SUCCESS in 33m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b6ac62ac90b049c6af876942580b0752 : SUCCESS in 1h 12m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a14ca099d5af42f2bd60dc03046b2756 : SUCCESS in 1h 18m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c3db94a1b9e04cc2a0b4ae70d7542f27 : SUCCESS in 38m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57dd453d39f84300bc11f7665525b919 : SUCCESS in 1h 20m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1683dd2e4ac54479891fcea3aae85a0e : SUCCESS in 1h 37m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f2cb4ee1a15490bb718a22258b91a9c : SUCCESS in 1h 28m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/453f2edc671c413bb29af32d34b41c55 : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":8},{"id":"812f511e74fad44597ad3bb0828b8b3a96e9c900","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-12 16:33:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a29a8a69a27e43acad09ea254bfdb6ff : SUCCESS in 1h 12m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f05a1edb234a4413a01b01f46e735267 : SUCCESS in 1h 20m 34s","accounts_in_message":[],"_revision_number":8},{"id":"cea24c2d5c2fe224af1e8d52d363b724bd583409","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-12 18:01:44.000000000","message":"Patch Set 8:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/17/913217/8/check/fujitsu-eternus-dx-iscsi/4aa80fd : SUCCESS in 2h 29m 44s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/17/913217/8/check/fujitsu-eternus-dx-fc/51adaa6 : SUCCESS in 2h 43m 06s","accounts_in_message":[],"_revision_number":8},{"id":"0adc3dd74e03a4687ab6ebd0cd662beba82dd313","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-12 19:16:12.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/913217/8/check/DellEMC_Unity_FC/90acc88/DellEMC_Unity_FC/2217 : SUCCESS in 1h 31m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/913217/8/check/DellEMC_Unity_iSCSI/bd726ef/DellEMC_Unity_iSCSI/2226 : SUCCESS in 1h 48m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"83420c512707e128e6aaf6fd21e71c34b45367f0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-12 19:42:33.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/17/913217/8/check/DellEMC_PowerFlex_3.6/4f0af05/DellEMC_PowerFlex_3.6/1702 : SUCCESS in 2h 09m 35s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/17/913217/8/check/DellEMC_PowerFlex_4.5/9699570/DellEMC_PowerFlex_4.5/583 : SUCCESS in 1h 57m 54s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/17/913217/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/1869f2f/DellEMC_PowerFlex_4.x_NVMeTCP/537 : FAILURE in 12m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"b102771698e348d1684dd65826f152eaba89d52d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-12 21:01:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/226/.","accounts_in_message":[],"_revision_number":8},{"id":"1bc7387bf11004fc8277b7aa21fd84fe96d72d7c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-15 08:19:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/723/.","accounts_in_message":[],"_revision_number":8},{"id":"2fb254979e050b2d7b86233dfb75ead32cf67424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-07 07:43:37.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"76e702562f3dad4a42faeef456e476e777cae592","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-07 07:43:45.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/237/","accounts_in_message":[],"_revision_number":9},{"id":"6e10b58430ef5edb0fcbdb720dae34c06138bd07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-07 07:43:47.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/237/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"91f276d5769b7e41e38022a862150abd19e2a38d","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-07 07:45:42.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"e91fe9b017df449d12056e9021f8cf19da37545c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-07 08:10:58.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4a583108c60d45f4b61b877f25487926\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bf6adc2eebbd4af3a0f12c0dbc9da97b : POST_FAILURE in 20m 56s","accounts_in_message":[],"_revision_number":9},{"id":"1097c7696ac0c99caae879847f373ec7776cd0ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 08:11:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c2a8a868e649e7a7df800169cc7507\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c9f475c2f6f4c16a18465653005e98d : FAILURE in 25m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/29b48e0d920b41b2a55babb077aa580e : FAILURE in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1c4f22d171c40446e0449d091dd18c6fcc8fdee9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-07 08:22:38.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 35m 20s","accounts_in_message":[],"_revision_number":9},{"id":"fa866a9b46cdbde1c02bce0e75ec42e2a599313e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-07 08:29:51.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 35m 02s","accounts_in_message":[],"_revision_number":9},{"id":"d1e41abaad7d74e08f582127101433b5177e3fe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 10:18:28.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a25ff7d431940ada2126f5dd1f21145\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538b985b0d55488fb0c80d9e0e3c70c1 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/652fa8fbfd874d9dad27d55ca47b2ce3 : FAILURE in 16m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a8299ae855b427cbd3af61c9890a7eb : FAILURE in 14m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0b1189202de46fd9a2a2fb1d4b45be3 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c85579d2a5c4ca29f03a0e55784062e : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/da0d09e967ef47b5bed2b83a8c25d956 : SUCCESS in 1h 02m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/244fbd484ce94faa8b7a3c65e3415344 : SUCCESS in 1h 38m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c22381a33e0f4ac1a14268d727a52f60 : SUCCESS in 59m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c010ec924ea342daa9d98f119d166cba : SUCCESS in 20m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/930e8f6c1cdb426c8bbd3d7ec28f6c2d : FAILURE in 33m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c46714b5bae64446a1a699d9724f53fc : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81dfb0b3a508401a9ead00689b1c26e7 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8d552dc42e34f8da8092adc62d0b4e9 : SUCCESS in 5m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/906e51f1c2184fc2aa7d78005adc7ae5 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e8a984a8f5b54fc4949a6b22e4ab76ac : FAILURE in 33m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e8ee428d0c142c48e8403eca4f6cb5a : SUCCESS in 7m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f7572ecbe20a4b17b847a5e641804318 : SUCCESS in 1h 40m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b9d658959574f3bb55c3a7b32fe8b78 : SUCCESS in 2h 19m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4b369260a884f29b953a75b143bfe82 : SUCCESS in 1h 41m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/819354b4da2b42e2b9ff12eaa840844f : SUCCESS in 1h 44m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4cfe90ed9f954d2190026de4430d77e3 : SUCCESS in 28m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6be66649729e41ca81ae835928bda848 : SUCCESS in 1h 08m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61c35041a74a4dfda81c0d36dbf4aef2 : SUCCESS in 1h 25m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1dfc9c3f71e043ab8ab8c8de12c909e5 : SUCCESS in 28m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c57e2f9fc859495fbc977ff5d0addacd : SUCCESS in 1h 05m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1908abe6060c4810959b76dc414f3774 : SUCCESS in 1h 26m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3781e5393d654c7e921b6a19d23d4470 : SUCCESS in 1h 11m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b027a1ff341e4721a1ad59c3ce0273ef : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":9},{"id":"a0b82efda298c66b3d4b408ab2c903955f74c68f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-07 11:59:22.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/434b3d7afc3d4b9297be45ab47c473dc : FAILURE in 1h 17m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bee7711d135b4b5db4dc9b2d823f9f36 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":9},{"id":"f4ffc959452fefe9650a5c1a08c2797a47c3ff75","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-07 14:32:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4650d40321ea78150ebde81d1db9f6ab49f5edfd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-07 15:23:02.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ca93be6ca3104dcfa1f6b7d326f43fbd\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/919c0a347a424cb9951261d8c6c0b7cc : POST_FAILURE in 27m 06s","accounts_in_message":[],"_revision_number":9},{"id":"bcfd8c0aa06526fc98c221b16640eebf9fc08c32","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-07 15:56:57.000000000","message":"Patch Set 9:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/17/913217/9/check/fujitsu-eternus-dx-iscsi/d29da49 : SUCCESS in 2h 28m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/17/913217/9/check/fujitsu-eternus-dx-fc/02f8e08 : SUCCESS in 2h 43m 04s","accounts_in_message":[],"_revision_number":9},{"id":"f3a508ce0952f3b49e91f7981f1d54b3e36160c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 17:03:37.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b0c670979b14dc79b058a3ba2753f18\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e7600c68754f2ebf4948ac89c0982c : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9804126022fb427caa945706d1071cc8 : SUCCESS in 12m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13c744a78d5648e58cbeb6b17822e3c2 : TIMED_OUT in 51m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7931e0b5b224546b4d002385238a1cf : SUCCESS in 17m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667f3da10447442a8c82c6e212989d4e : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/480c79d8bc534f9b87068e4244671f52 : SUCCESS in 56m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68892950e9b547169e96d22ad9522aeb : SUCCESS in 1h 20m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/313d1c871ac04a54afac8c9f295ab36b : SUCCESS in 56m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/151442538d6c4e5986182a1a2577f0f3 : SUCCESS in 19m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d0c23f81b91b4a40af6014ceb7ba43ad : TIMED_OUT in 1h 28m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0fa3cc1f02c8448ea121bcc0e6bb0405 : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/493c6c99f9c6456a88f4ef49600ca521 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be725004206e4be69c58a17d681b7fa5 : SUCCESS in 6m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9a2592036ad94c34bf96346cb7754365 : SUCCESS in 9m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/65ee0c06a1d44df58421d6eeba354454 : FAILURE in 37m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c5f3518af1a4be382cf489b528bf507 : SUCCESS in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7b5981317b4b408bb4c4d9e1a34ec09f : SUCCESS in 1h 52m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d7cef32e378c44779f0954580fe4e48c : SUCCESS in 2h 19m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f2c0a7d066e64083a0c6ff9ca65b8f04 : SUCCESS in 1h 32m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/486b32e2653d42eba403941b950a3ee9 : SUCCESS in 1h 43m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f5f3fe7059b64a08b7535041d2420cb0 : SUCCESS in 24m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3493619e055f43b6a6a6b26150292e23 : SUCCESS in 1h 01m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/096e2f25948e4825a466dddc8865ae35 : SUCCESS in 1h 30m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9e810de4a19d470bbdb927ad4cbbfc46 : SUCCESS in 37m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/63c4e8711be04dc2887a711e85e99808 : SUCCESS in 1h 25m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dd54a7dc7662442aa94515b4467f1d77 : POST_FAILURE in 2h 04m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c59ded5f751442a9a0e1ad1aed0f208 : SUCCESS in 1h 34m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51ff0aa87a4247789f91fa9e28a726d6 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":9},{"id":"55351d5c109da1a7d23347fe124a1a46d33cee3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 17:33:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/787449424de9421b8424424df1de1720\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/117e85092544447482e70c4389e3773b : FAILURE in 30m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c9ccf8cccfe4416a8373ec91f050be09 : SUCCESS in 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5c50e641c1bc62af74785a8046232f40d7b375fe","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-07 20:13:45.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/17/913217/9/check/DellEMC_PowerFlex_3.6/5a63a59/DellEMC_PowerFlex_3.6/1864 : SUCCESS in 1h 54m 24s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/17/913217/9/check/DellEMC_PowerFlex_4.5/23fc9da/DellEMC_PowerFlex_4.5/743 : SUCCESS in 1h 49m 11s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/17/913217/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/2673290/DellEMC_PowerFlex_4.x_NVMeTCP/708 : FAILURE in 14m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"a01ee8818b628b693255eedf47008943a7d7c46e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-07 20:24:48.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/913217/9/check/DellEMC_Unity_FC/ce42145/DellEMC_Unity_FC/2378 : SUCCESS in 1h 45m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/913217/9/check/DellEMC_Unity_iSCSI/d7b15d4/DellEMC_Unity_iSCSI/2383 : SUCCESS in 2h 21m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"6e6d38bc2a6c6d8e11b8ff24a15ece7b1611dcce","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-07 20:30:32.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d33b665f2c3e46fe9928fedbd6b78119 : SUCCESS in 1h 11m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/09104fcd0b1242309a127dd4961e9f1d : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":9},{"id":"69b732d420f679d0f6cdcff8a7fa39bc541c74e5","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-08 02:38:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"695d78be60a151bec9f8ebbd6bb3b0ee35116656","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 03:18:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a62b76516dc54f69a32106156c2ee4e0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e247738cefe49918502ed764cfd9e5f : FAILURE in 35m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5132f9255dec457390a0647468b5f7b8 : FAILURE in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b0a0452190c02653272dc180663278180bbe4721","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-08 04:23:13.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/9/20240808091542/NEC-V-FC : SUCCESS in 4h 05m 47s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/9/20240808091542/NEC-V-ISCSI : SUCCESS in 4h 07m 27s","accounts_in_message":[],"_revision_number":9},{"id":"42375ac2251f8c23e5403cc199057dbf1868af44","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-08 04:31:29.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9f4fb6405cff4917af8cd1ecc34921d9 : SUCCESS in 1h 34m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/800c3c4a95174ec499c3af671d257cd7 : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":9},{"id":"d82b436e7470c7a7bd7737a248e8d0437f93043e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 05:22:12.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8fe111798fb4201afb83dd8478d4e36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d302812b36604c488c940904d01e9086 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47caf12f344b4a88b0275ad0bed3bf91 : TIMED_OUT in 1h 33m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/385e5a3c2dd04c4f8c3d9b3d1cbb3a96 : SUCCESS in 10m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46f15159ca344b9091cbaeb79dd7d8f4 : FAILURE in 33m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387d8ca1252945f1929903eb13f7451a : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/d2527729590645f2a21dbff7a9816a47 : SUCCESS in 54m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3b51b5ef81545d3996b0b0fc417c955 : SUCCESS in 1h 31m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b21e51992b3f45e9a9034e63a5ecf453 : SUCCESS in 56m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4982eee6b7bb40c39a2dc64c1a51cf2d : SUCCESS in 22m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fddaa69576d4412b8ddfdf1a51625023 : FAILURE in 32m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e4f32e89b2174704b59b1560c31ec757 : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/68af17f7c56f4b0d8b1ee002d5329785 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9fafbdf099cc4c26865cd1ce2d923a66 : SUCCESS in 6m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2d386ff46828485783475a1b6eb8c185 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fcdb02d1ed2a49069a4321d48a55f4af : FAILURE in 35m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/23813a123ef14d57a20468edf4057ae9 : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd7b3635a7c04d269e859f49000be09d : SUCCESS in 1h 47m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd1423ee8b3f46249d81544afec2a12d : FAILURE in 2h 28m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3af7f37667d7468aade53654b629c7a1 : FAILURE in 1h 53m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6fbf63ad9193468a80ce601d55d537f2 : SUCCESS in 1h 49m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cd28c3f1852c4d7aa5edd4a8bd62d161 : SUCCESS in 25m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dc4401b3b6a947adbe483331ed54d5de : SUCCESS in 1h 04m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/15d2104a4313487f9c653cc02a13612e : SUCCESS in 1h 19m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86911f78ab4d4003818121e3d4686bc5 : SUCCESS in 41m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0edcf3fc68094f51abc7002543a4d54a : SUCCESS in 1h 05m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9bae6c6bd3704dc8ad06f774016f3083 : SUCCESS in 1h 29m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6d004158057941078bf887474282cd04 : SUCCESS in 1h 29m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/966c9d1a9f4247d8a31051e335b1f2a0 : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":9},{"id":"f5ea9d058ad84031b9180f7e9013297fbae80e16","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-09 01:06:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"29ee101eaec9ae033fcb1914287f5b158e1184ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 01:43:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd02ff4c23f49a1b56bfff9579463d7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7207b9ab56f5438894024196a81f8a12 : FAILURE in 34m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f0e0e0670f5343cca1db13635c34975a : FAILURE in 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"56bd2c5b6ddc36ef5aafc6123a6a70624377ba5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 03:20:23.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5cbcc38cea84605a38a47086b534c4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd7792e477441bcad2637168863a3a6 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab8017b3aa1b4e11a6d13b6945a10f5f : TIMED_OUT in 41m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a85bd7664bd5460fb7e2b98c6480004e : FAILURE in 9m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/390598e40fd8486f8bfeab8b32b40ec0 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b31513ff20b4ff5b873f55c35ed88f9 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/4b170915bd424afeaa66f4dcb7bb7073 : SUCCESS in 51m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d5fe3357ee3467da3eb03bc9e99c704 : SUCCESS in 1h 23m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cd7f098ee0f49b1a88c436a21ebf0ff : SUCCESS in 44m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ad4ba35d9ae43c39fd467de1b778cc9 : SUCCESS in 15m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e3394d9d328545b29afb5158afa5083c : FAILURE in 20m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7fda4fcdc38a4aab88e2db03e499a91f : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8be28410e58476bac9ef0c99a75fcb2 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b6ac15bffc142d1b038b897bef06fb9 : SUCCESS in 6m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1f91d5dd427e4a3f9879169355cc2785 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2cf04f0d17324fbe8d029d9549f5d825 : FAILURE in 26m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e112f6daf8cc4d5b91dbd584e41391a3 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bb6ec83f0ecd4d4dbf5dc60488dbcdb0 : SUCCESS in 1h 49m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c94817563d494061a7a574aadc44d5f9 : FAILURE in 2h 10m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0ad5d8599d6a4301bca051a32b25f7a4 : SUCCESS in 1h 51m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b153ee3b886c4e4cb8c59f52f2103adf : FAILURE in 1h 39m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cf4f2a41e40246b1a5e94cadfa9c0cb5 : SUCCESS in 37m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8262e0c753e84864b34206680e465f71 : SUCCESS in 1h 05m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ef919a33dc54f7baaf7d7b88f085e7b : SUCCESS in 1h 27m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a3b016bc919049eca370292de7a63b81 : SUCCESS in 29m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/30efcd2b8fad4e8cbb6fde2803c4fee4 : SUCCESS in 1h 30m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/19d123d3aa0c484db9f7b304c75f6f11 : SUCCESS in 1h 34m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe898395cb3d472b879b43a58e160239 : SUCCESS in 1h 34m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89ee77be215049bca9a788d13544b13b : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":9},{"id":"69b5af7e31ae7ec9ed6702d5124dab34e9ee6297","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-09 06:27:40.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5dd50cc5f64b4ae98c51779fb19ce9e7 : SUCCESS in 1h 24m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6a0cddb5584c4775bfdc48f4d5bcacbf : SUCCESS in 1h 20m 11s","accounts_in_message":[],"_revision_number":9},{"id":"b101bf60daaf3dfbce6c29b9e820a132f5c00189","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-09 14:18:43.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/9/20240809143927/NEC-ISCSI : SUCCESS in 5h 16m 16s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/913217/9/20240809143927/NEC-FC : SUCCESS in 8h 39m 13s","accounts_in_message":[],"_revision_number":9},{"id":"00b0715cebcfbb2cbaaa20ddef7137462ca18fe5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-12 08:22:08.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/862/.","accounts_in_message":[],"_revision_number":9},{"id":"9256a4ed27db6e649a856e0b1a4ac13eb1e292b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-09-04 03:40:18.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"c727af7f5a2fb881ce52fbb94016b6468fe5f521","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-04 03:40:27.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/476/","accounts_in_message":[],"_revision_number":10},{"id":"dcde269830e55d810372521b356dfd4845aa3ee6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-04 03:40:32.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913217/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913217/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"dcd7611b94d93fa5b12680926f627cc523ef7678","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-04 03:57:07.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 11m 42s","accounts_in_message":[],"_revision_number":10},{"id":"d3e28bbb1acbbbd954f257a702beb8d362e4b925","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-04 03:57:19.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913217 : FAILURE in 11m 42s","accounts_in_message":[],"_revision_number":10},{"id":"0d81ddd7424302d1f9334de4cbe122ddaeb79bef","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-04 03:58:02.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 12m 16s","accounts_in_message":[],"_revision_number":10},{"id":"4efd6545732b7f66d268866f41c9cd47013ed70b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-04 04:32:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/479/.","accounts_in_message":[],"_revision_number":10},{"id":"801b1d525bdd7c867202751a9c09050d07da3aad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 04:52:18.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e9b217d531041368ca0a4f4c249740e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e52f59c47114da0a7f8e8ab19c6222a : FAILURE in 22m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/980affc599734248a678601a35bb403e : FAILURE in 21m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bd532806527f546685b4300210dced341cb65419","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-04 04:54:55.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1db62eea11fb43359a5aae984893a58e : SUCCESS in 1h 08m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2c8e8ca976064c83b6c9c846e28bf42c : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":10},{"id":"9ceb76e1328db5e8526f43e030da189f896e7603","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-04 05:09:53.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1067/.","accounts_in_message":[],"_revision_number":10},{"id":"7ee35a358cd0403b256a7de05f0dd1948d4ccfa3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-04 05:21:27.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c0eae8838cdd4561b43c962387a718e0\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3c80991755914fc790c64d7118997b8a : SUCCESS in 42m 58s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a106aafe064f466196e66edd284bb9fd : SUCCESS in 39m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/87604a30a3c34ed7a1f9bdb3eb1f2828 : FAILURE in 1h 39m 17s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9afb5fe3363642359b21a13a13830302 : SUCCESS in 42m 11s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"a1622d97b8be5f9cc33b09dfb8ddc9c25fdaee87","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-04 05:48:02.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/913217/10/check/DellEMC_Unity_FC/fe4b9f2/DellEMC_Unity_FC/2547 : SUCCESS in 2h 04m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/913217/10/check/DellEMC_Unity_iSCSI/255cfe1/DellEMC_Unity_iSCSI/2552 : SUCCESS in 1h 55m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"17255c9a86b4f4650aa96849b6be454e7cfaae1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 06:01:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/211729a6ed3248e6a1719cc767739f87\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffec37f7a26479d8d63299c90f1daa9 : FAILURE in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/037246f4ef164a8195cd638703dbc002 : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6488d3dd61f74af8b02827113e101fb9 : SUCCESS in 13m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1776207848dc48b98205acf9a6a157cd : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b47cff1b9ab4fbb8dc5497d8d7675dc : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/64cf9844fe884096a8023e78d7080712 : SUCCESS in 1h 01m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee8fc47b66624ffdb82e8112e63cb9d2 : SUCCESS in 1h 39m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39c9f505be9d488dad65b5f401b26f79 : SUCCESS in 53m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c48e017822244ebb1ab5b3fc06271d7 : SUCCESS in 15m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7ac83ec61a7244c88ce2fc48a2d8300a : FAILURE in 38m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7d8d9ecb252b42419edd68e99411a7b1 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef7a032e8978409cb6ba1cd370d4a27b : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b73ba4fb1804e4e80b7cbb2ba704ca1 : SUCCESS in 6m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1ce50aba0b1c4c12a97890c6c10bd2f0 : SUCCESS in 7m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b440baae493d4a39b12dd36948abaf7b : FAILURE in 36m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad16b56b366d42b79a0f78dab2ec6fa2 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/904faade53e54f5ead17cf73d44e996e : SUCCESS in 1h 59m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2da4efc8c4be459ead6fb0e611ccf96c : FAILURE in 2h 15m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0e6d20e442154cccaff09306f81d08cc : FAILURE in 2h 06m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f380040dc3f74f61b87c6b53f024d813 : SUCCESS in 2h 04m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d065b8cfaf804033bb35d2d3bcb8a0f4 : SUCCESS in 27m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/50e43596f38143bcb4099685c33b971a : SUCCESS in 56m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/427384cf3a7d49aaad4b125f11ba0ec5 : SUCCESS in 1h 29m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5e505c24b6a542d7956b672c7752e22d : SUCCESS in 32m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b334004dee1f453882b45e61275a0d67 : SUCCESS in 1h 30m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f60cef499af64b608a0b90878843be76 : SUCCESS in 1h 34m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51a3eb8662584e8f95f8157de456a61b : SUCCESS in 1h 12m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7b8e422e8ac40d988caece188fb5458 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":10},{"id":"20865fe6f9d56cea4c01a32a8dd1b0075a6fbfc9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-09-04 06:09:05.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/17/913217/10/check/fujitsu-eternus-dx-iscsi/fa55362 : FAILURE in 2h 20m 37s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/17/913217/10/check/fujitsu-eternus-dx-fc/5db1621 : FAILURE in 2h 23m 39s","accounts_in_message":[],"_revision_number":10},{"id":"840102c71f4b765753674a01a474c90c62d7f9b6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-05 17:43:21.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/10/20240905163216/NEC-ISCSI : SUCCESS in 8h 48m 28s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/10/20240905163216/NEC-FC : FAILURE in 10h 11m 02s","accounts_in_message":[],"_revision_number":10},{"id":"10c0a893c6e24dcccca094811d1c3ecad8f4a040","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-05 18:13:08.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/17/913217/10/check/DellEMC_PowerFlex_3.6/36c64cb/DellEMC_PowerFlex_3.6/2043 : SUCCESS in 2h 18m 30s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/17/913217/10/check/DellEMC_PowerFlex_4.x/b32b6f2/DellEMC_PowerFlex_4.x/926 : SUCCESS in 1h 37m 14s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/17/913217/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/0a27d79/DellEMC_PowerFlex_4.x_NVMeTCP/885 : FAILURE in 5h 53m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"0041f85f2b8824ec73e6b54d434c0b0b0c7c0607","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-05 23:12:09.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/10/20240906000526/NEC-V-ISCSI : SUCCESS in 3h 58m 26s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/10/20240906000526/NEC-V-FC : FAILURE in 8h 06m 40s","accounts_in_message":[],"_revision_number":10},{"id":"577cbecedef62eba90abf4d14330be8c120dbdb9","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-09-06 09:14:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"74c1ed3c3cd57ad8a34aa971c44ff69fa879536f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 10:11:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa9bf53d74541c48d51e63636ba67fc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34124fa849154f9983c5b287a2f8d619 : FAILURE in 26m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f42f94167c4c434eaf2c70de362db637 : FAILURE in 20m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"25ed1128d9296810319c3fb23f05a047ece2dbdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 11:52:27.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3c81ec770804fcd82f9478150e7e4a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc794b0332847c0950562c3fed274a4 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcd1719069324e0687fde937494644d2 : TIMED_OUT in 53m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/876249dce0014bddbf9c168bd8e46c0d : SUCCESS in 12m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1caed3eb5c34c95bb9165dfc0d5f140 : SUCCESS in 13m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af04ebcf3b90480ba516a9eb03fe25fb : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/153ddd241898429f8e036d3756546327 : SUCCESS in 1h 05m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6faf3c96868944ce80cad0b09ff0975d : SUCCESS in 1h 43m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4461a79aa8e94b138e59e54627aa5fbc : SUCCESS in 1h 02m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0622957f798f441b9f306547fada06a0 : SUCCESS in 20m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/69c99ef6036c4651aa4b4e95e6a84da1 : TIMED_OUT in 1h 39m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c0210d81e49b41d49bb7a7b4a094d0df : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0355e3f344424e6aa2a3b4ba01880ca1 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b38e8487357446a9a4faca488ec5519 : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9bd110f19f1d4adeb3c88cd72034a031 : SUCCESS in 9m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c5b1427a7905445db1c400103417e4fa : FAILURE in 35m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88167680cabb486f8086809ea125e50f : SUCCESS in 8m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6aa0f225799c4ebdb515f51317230375 : SUCCESS in 1h 53m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/265e7e5b354e484a8985e6e34f2a6535 : FAILURE in 2h 29m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4f655f5b525649f6b9ce7fc39cac04ae : SUCCESS in 1h 49m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c58b412030d3405eb6c2f7ff02ad3dec : SUCCESS in 2h 03m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b544cbbbe7154589934569f296ea3f9e : SUCCESS in 36m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ec0c6deda18d447a93f8669af41af337 : SUCCESS in 1h 19m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a08d31ad57d6476c843551b6e9187e41 : FAILURE in 1h 46m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/47bf2b8b92964b97af35ae0cfb24019f : SUCCESS in 41m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/58ee47686dc744fc93b833de4aefd86f : SUCCESS in 1h 29m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0fddfe9cfde74f32a3dc76e621a96f81 : POST_FAILURE in 2h 22m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b5921798b92d4245b1b309778746c646 : SUCCESS in 1h 50m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3bca5bcef7c49d392450880a3e74ffe : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":10},{"id":"6c0a9ece5da56c0dc84e0306afc7f3b9488a0846","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-06 14:25:38.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3aa694cc94fd4382a01566471f6bdf5c : FAILURE in 1h 08m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e8fb923b7a844fccb63ec53434cf4115 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":10},{"id":"71f6f3300cb02129fdef4daee598b669e512517a","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-11-04 02:28:08.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3bfb2622e3a5cf22b398a98a4cdb2b7ce0870cd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 02:57:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7fa5aced7045eb9d33b05cceeeca97\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8860325ec364cca8bf741ef1bf58637 : FAILURE in 24m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6f4180a0470345959a69c6fda18cace9 : FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a4955a2e6fe6ba12cbfe6594bd57dfb3d73bdafc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-04 03:51:59.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/162fcf6081e34c0fbec7417705a24a3a : FAILURE in 1h 20m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2a9586562c574877b80e2bb4cdde7fd7 : SUCCESS in 1h 17m 57s","accounts_in_message":[],"_revision_number":10},{"id":"9ec39e60fbebfbae1748ba96abbbdbaf497824a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 04:50:19.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87abfab3bd294fcf9902b925d38978ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e55e99ec0e8d4c338fb777384c2ce7b4 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c52364df12a4a87b216890d65a1dd18 : SUCCESS in 11m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc4245f80d944a3b833451c1c64acbb5 : FAILURE in 16m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e912c42f4c42448423f55af71c8066 : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/38bd8e4797e441ac8f6961d29c887e42 : SUCCESS in 56m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/96ab2ef142f748d1ab81d9abbd9fb923 : SUCCESS in 42m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80ef4f34a4ca421abeaa3640ac31c078 : SUCCESS in 1h 33m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4dfc2ee795141b38c82a82ab862ccb4 : SUCCESS in 1h 07m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db7c4b09d1aa4f78b1c42803e107305a : SUCCESS in 17m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d06366046ec4b0e8ae3c1f6b795c97f : FAILURE in 27m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/450266e26f50437c9d8a9fbeaf771f78 : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f2509b13f7346eca6b045706b387712 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a2a59264a1cd4c6aaffca31755eb7088 : SUCCESS in 5m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0561b89c1053482292ca7aecb56f385d : SUCCESS in 7m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fa6402883d7e42b29b8905c3eadc7c09 : FAILURE in 19m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b064143a8154948a705a6085ccb6a3f : SUCCESS in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/60d98d12977a43b8995c3abcbf90814c : SUCCESS in 1h 48m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1a26d35a3a4f42328b06ef4573db2280 : FAILURE in 2h 12m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94dd9a2297774ee7afa290ab3f44955e : SUCCESS in 1h 28m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8b16d5a9dd4d446697553ca782f90d6a : SUCCESS in 2h 00m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6b71a84f46e94e69b99b520bbfc23157 : SUCCESS in 30m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d73be1cc517747f7948b2a890510111c : SUCCESS in 1h 06m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8052ff55073a4aa2a5bd5af532d8ab85 : SUCCESS in 1h 20m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7d29ec3100ab46bc9d42783c15bfe131 : SUCCESS in 31m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b480c2f816b4916adc671142a1c90c5 : SUCCESS in 1h 39m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d57188519bbe4f52ba9892cb947221ed : SUCCESS in 1h 27m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78d544a308f045c49a01b9c491f007a6 : SUCCESS in 1h 42m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66f81739562d4fae9dac5f6b1f39bd9b : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":10},{"id":"d6474096b081d5bbacf3f9bc45043325ef77e7bc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-04 05:00:09.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/17/913217/10/fc : SUCCESS in 0h 52m 43s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/17/913217/10/iscsi : SUCCESS in 0h 56m 44s","accounts_in_message":[],"_revision_number":10},{"id":"3d9c86154870c1402cbbaef0fccaa75477b39d1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-07 06:27:21.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"97eb35f39e0a493f5d92b445a2e4b6fa304bdbb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 06:27:32.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1155/","accounts_in_message":[],"_revision_number":11},{"id":"7832ed886cb9c0752e0077481617e57596218fc0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 06:28:24.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913217/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913217/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"377e7fdb5b3ff0366b81a88928ebded471a36417","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 07:10:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca130ee7a0c54531b7b99f0ecdca5259\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb635f7760af41149257b67962d14d24 : FAILURE in 25m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e501eb1e2884482a8bd829b6bf0021d : FAILURE in 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7b3a3f98790b71e5bac0c3aabf0102bdc30255ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 08:45:59.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b902bea8ffb4fdf87c8b835dc7f91a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b95730764b4a1a87b55fee54ac3e1b : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4db50a073214f86abc5b5446f281b6d : TIMED_OUT in 42m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2457ee02a1664dcfa41d7e977b275332 : FAILURE in 27m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbe00059ba54ff5aae6c91c66dba8ab : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/f4b85f3bc17143f1b6198728bd0f30fb : SUCCESS in 1h 09m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/095340ec909447af950387e0584984d6 : SUCCESS in 1h 00m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a519c64286ae444cb21eb1e2a1013ca5 : SUCCESS in 1h 37m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca0e981047204a73a3377558175cf0ae : SUCCESS in 42m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b97ee1536604c1bb5835f46f0bb9214 : SUCCESS in 17m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/12f0bf319fdc464397cd97a458e988a1 : TIMED_OUT in 43m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/db7f20b29c074eb88befd7019aac77e0 : SUCCESS in 6m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a227865b6c4b45d282fef96da487a4ff : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/302af9ae22264a7ebafca3159f3d41f1 : SUCCESS in 6m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4d3369bc7fbc46f5b4dc32b44e2ba866 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a939899f31404d098176f92665c1b697 : FAILURE in 21m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e22a3527046448e8bc1703e65c11114 : SUCCESS in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f64785f6d2d04828b44882a9f7f171f1 : SUCCESS in 1h 56m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/585ce90162b04af89e3faa9bb1895d2d : FAILURE in 2h 09m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c0a1de91ab0a4902aacfe127c8c62c12 : SUCCESS in 1h 45m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1fb7ebb14c094c73bf1e49d408505f1d : SUCCESS in 2h 07m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2cdedbf04ffd414dbc6b4ed216f01253 : SUCCESS in 31m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b6688876045e48b9863f37b3e89f918b : SUCCESS in 1h 15m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cf93d268c0fa4f889bf0163abdc3615f : SUCCESS in 1h 38m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a7f02222cb0549e0bca9e0f89a866609 : SUCCESS in 34m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cc48affa16554618884918dcef12595e : POST_FAILURE in 1h 34m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d1e83e0fbf3545d8a7eb36a5c47d7812 : SUCCESS in 1h 32m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/65b4e569d5234bb6a885ff5decf7d5a3 : SUCCESS in 1h 50m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/106d6268874a4f0e94d87f7429165af6 : SUCCESS in 1h 16m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58be40b0b2fa4e4694f36bf43c840a4f : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":11},{"id":"86fce566526ddcf5d0b17587a707e23587d23e8b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-07 09:32:09.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/519ee77948a3466487d21066f69066bd\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0746042d43684b61b2e5d3de7b4bd93e : SUCCESS in 42m 58s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1a454158b8a47b08cf7a479cf18b8f2 : FAILURE in 47m 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/866dd483d4c74ae4a6bfd2cbf8539042 : FAILURE in 1h 40m 06s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc24703ee59e470b90fc3e47cc8984f0 : FAILURE in 42m 52s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cf6e351bdc224a17bea01a248ecfe57e : RETRY_LIMIT in 51s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9448010d9da74014a8ddf2f6c9de2b8f : NODE_FAILURE Node request 200-0000021192 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c899676711f45d4ab7497e3f71716ec : RETRY_LIMIT in 54s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/746baffc073545619d32b7d1183d055b : RETRY_LIMIT in 46s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"595c3d4e408d8aed1353038341982c59b2faa9ac","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-07 09:50:09.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/913217 : FAILURE in 34m 25s","accounts_in_message":[],"_revision_number":11},{"id":"bfab129a6cfcf004818774c6d4278d8acf31ffd3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-07 10:10:31.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 35m 52s","accounts_in_message":[],"_revision_number":11},{"id":"2119cfee52c6f1cb98a749fbe29b91bfba29d1c7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-07 10:17:21.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 35m 59s","accounts_in_message":[],"_revision_number":11},{"id":"3ac88c7a7198d67da7214a04ad7ee05bd8dd8bfa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 12:38:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/193/.","accounts_in_message":[],"_revision_number":11},{"id":"66b36ea7cc9e42e0b339cca828743fcd2b278250","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 14:00:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/182/.","accounts_in_message":[],"_revision_number":11},{"id":"ba6d4a499a76d15b588db9dd7302d6f83cb15ebc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-07 15:59:31.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1f0a194e5a2949f9a6b31a7e682a7f2c : FAILURE in 1h 19m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c883515e84544fe9acc738e3cb436ed9 : SUCCESS in 1h 28m 54s","accounts_in_message":[],"_revision_number":11},{"id":"41a09c5758316fe25a4554b58225e1401fd2e36c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-07 23:25:33.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/17/913217/11/check/fujitsu-eternus-dx-fc/44e742b : SUCCESS in 2h 29m 49s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/17/913217/11/check/fujitsu-eternus-dx-iscsi/bc5a36f : TIMED_OUT in 3h 35m 44s","accounts_in_message":[],"_revision_number":11},{"id":"b066d527c82e236e78755420673456a242121117","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-08 04:29:40.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/17/913217/11/ : FAILURE in 1h 24m 04s","accounts_in_message":[],"_revision_number":11},{"id":"139b657777586e6cdeb1d6c9b12ef0a6859e4fce","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-08 06:01:53.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"13f63c61398e98ebda599be0e6797a2e1e5bdb4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 06:33:32.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48efb4a1634f4bbfaf89908bb81913e6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d00ab21c9ac44d3b847615862c468534 : FAILURE in 24m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb6c2e14111d4a809717df953ff7be53 : FAILURE in 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ba82a63c73727acc9d7bc06f66d4c525ed01f082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 08:18:43.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35074b80d5b94c31b9330ee335ce95d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9da1fb4e05ce43109988b194bf19c987 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0354274e9b6f43a88b4b01bcd9e38daf : TIMED_OUT in 1h 28m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/244cdec250844ce8b4a2be7ed1786076 : FAILURE in 28m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aecf7f38a2134b7e931cc2aaabeeba0a : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/d2cad36cc1bb4c8f9c5dbc5b274352d4 : SUCCESS in 1h 11m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4e3ecb3f145e42be88da108ffba5f175 : SUCCESS in 1h 07m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d77a89aabbcc452dab7d4d81607f57d9 : SUCCESS in 1h 40m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4865d4851404243965c63572986cc06 : SUCCESS in 1h 06m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d9e90bf6ec04ea6b5a52e7e2117ae2b : SUCCESS in 18m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/653ee006dafe449aa7a2e9114930ab82 : TIMED_OUT in 41m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9fa4a6d91aed4c158e8d7347f905531f : SUCCESS in 6m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4aaf2d2e997843468bfff523c6d8d003 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3f0277d9f8246bb84c9f37472a5979c : SUCCESS in 4m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f5ddeeb29cb147a5bf8437e78f84bd49 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bcec21184a13438aa02af5d6b279c222 : FAILURE in 27m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/25cbd60a86284f85aeb14af2daeb087c : SUCCESS in 6m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/49ce6e879b5a4bb98fdd22bf9a2c8300 : FAILURE in 1h 57m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6f2fc2ce4eca48c584ff115a90071c0f : SUCCESS in 2h 12m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c74664c67ea8452ca78bed6300abced2 : SUCCESS in 1h 55m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e54e356d3cb24fb9ba5517fc447b5f4b : SUCCESS in 1h 55m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0bde195940484b76b4a5ae34c8f62bab : SUCCESS in 27m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a48a52645483409cab2e169a09ff187f : SUCCESS in 1h 06m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/89bc501bf6bc4a668751caa66e64f945 : SUCCESS in 1h 25m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2afcff00f166415ebf4c3a397df60b63 : SUCCESS in 27m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78e16c78bf594ce49ea7b56bb340cd40 : SUCCESS in 1h 32m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0897405660144a44ad5ff653a04616bb : SUCCESS in 1h 13m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75f575e8f363420d812cda025eb93259 : SUCCESS in 1h 46m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d947460503844c6afc36883c7878410 : SUCCESS in 1h 06m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01caeeec4b2a4a1cbd5ff1c658ab8172 : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":11},{"id":"6e120760524ce971e87760ec3e8ca808d64120b8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-08 11:30:40.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cdce9b778a2f4cd781dabb6337f3c481 : SUCCESS in 1h 07m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/13db71d554084687b78009b1f9ef2516 : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":11},{"id":"24fd695bb1d543c9b1b12fb9be6db7e7c795f430","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-11 02:35:18.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/11/20250111040516/NEC-V-FC : SUCCESS in 4h 15m 16s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/11/20250111040516/NEC-V-ISCSI : FAILURE in 7h 29m 59s","accounts_in_message":[],"_revision_number":11},{"id":"bb9f08fb31f18d8d9a98903dda744efd2d6eaa09","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-12 05:40:35.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/11/20250112085141/NEC-ISCSI : SUCCESS in 5h 20m 44s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/11/20250112085141/NEC-FC : SUCCESS in 5h 48m 51s","accounts_in_message":[],"_revision_number":11},{"id":"fe9dbec21ffbd9b41b110bf0279916756d0bd4cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-21 07:53:29.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"b09c308c2c5be13cabe691ebd03f2cbac000e5b1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:53:41.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1549/","accounts_in_message":[],"_revision_number":12},{"id":"11e7301cfbf3bcd58b76c9b3ff87eeed7971ea46","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:53:43.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913217/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913217/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"3b95a40ffb83792fa8505d2576dd59b05c6bfafe","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-21 07:55:33.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"b605b379494e61805d8b5e615f93a3fcae2f85b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 08:22:10.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b20e9b8796441f1950263cbf2c8235a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e399f74e4a5465ea3b06da1ebf8bf16 : FAILURE in 26m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aebac31b613247968f01d4f302f496b5 : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"039869da73e045606bd0673761d94025a32b147d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:45:53.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eae1b40343334c8990fd1a77bcc6d9e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff13b9651834a4aba8858983f2a9ab2 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d038e43b1183483c861cd9d3e0f354ed : SUCCESS in 12m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5261e7d977b74a70872c5f8e8e8d624e : FAILURE in 15m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254a3dfabae5489b8a57483fb082dcc5 : SUCCESS in 15m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/5f84b02c98374ba1ac594a9a62504040 : SUCCESS in 53m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61e9f8a2449044baa40b0026ccb10b74 : SUCCESS in 1h 15m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/113ca206014343b4ba7a5d21e868eef3 : SUCCESS in 1h 47m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08fcb419963c4b98ac9fac6c21785ffb : SUCCESS in 1h 15m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78d6f775a5704ff6a5bc18491bd0ee03 : SUCCESS in 21m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b96d277213b74360bf93a96fdc9d4a69 : TIMED_OUT in 41m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/20bfc04f50bf4efebf0bd3492af7a484 : SUCCESS in 6m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b5175f80f2d41d69610024a7d47658d : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/89fabe62fcad423e9b693907410cbf19 : SUCCESS in 7m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c07f917d277a417aa9dbd2fd26b5e13b : SUCCESS in 4m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3af68ff81a7c43dbb747835fc98fd95e : FAILURE in 34m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33cab26c7e99489cab4e84f179c9973e : SUCCESS in 10m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a64b28ad261c4ac79265f737b095495b : SUCCESS in 1h 18m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/526c2c1e89594171aae8cd0d4e08aaa5 : SUCCESS in 2h 18m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40090fabcaa94e1380ab69630b9d3e64 : SUCCESS in 2h 20m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/80c83bc824364424a8960685a7cfc021 : SUCCESS in 2h 11m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2ecdf834014349a69d3d375e1ce80acb : SUCCESS in 55m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd2762e54e6e409bb9706207eb41e950 : SUCCESS in 1h 07m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5d4c6d1073ee475ea993ada1dd0e0eea : SUCCESS in 1h 29m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca40cc74246d4359a00c7495f06acc7d : SUCCESS in 41m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9f35a5ee13d440859f9a508fccddab5c : SUCCESS in 1h 37m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/792268c57a404488af76a3d7f0edc462 : SUCCESS in 1h 38m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a13215cd909543de9f8e9d21daa085b1 : SUCCESS in 2h 16m 32s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75cf719914f84a64a0b70d48d597dfbc : SUCCESS in 1h 26m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f5337e957a548e082964e39abc016ca : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":12},{"id":"b1b34a1234fcac0250fa2754f444cb952290026a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 10:56:42.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/486/.","accounts_in_message":[],"_revision_number":12},{"id":"9f0e1c7db97e74b9c34252aae3565f95af37d0eb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-21 12:17:02.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6502c4dc83764c7ca66302bdc19d8650\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b135d03854d49d3be052b69677e9776 : SUCCESS in 45m 02s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c6d2a6c2ab5649ec953e035649d7e3f2 : SUCCESS in 44m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec217e89f99547be9d161361d2655dbe : FAILURE in 1h 35m 38s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc8917646cc540a78b292b6a1af651ec : SUCCESS in 44m 40s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c4b47ebb24a444889a306af393523fd5 : RETRY_LIMIT in 40s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ccabba47c64e409da93980f05d4cb511 : NODE_FAILURE Node request 200-0000033815 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b9d60b1430a4af48e2d36bd46630653 : RETRY_LIMIT in 39s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4b19f25298df49fb9bc88a198aeabbb3 : RETRY_LIMIT in 42s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"b74f9c620cb1434663ead658e180622058ef31f1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-21 15:21:38.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 34m 43s","accounts_in_message":[],"_revision_number":12},{"id":"33cb928707520b55acc49873591ca1a5e5a83d9e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-21 15:25:06.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 34m 25s","accounts_in_message":[],"_revision_number":12},{"id":"24c08325a83a243326b0d0cd093394eeb744bdcc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 15:50:36.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/555/.","accounts_in_message":[],"_revision_number":12},{"id":"af153ea6c355b5eafb4983608bfb254a612218d6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-21 19:48:21.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-12/FC/ : FAILURE in 9h 49m 59s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-12/iSCSI/ : UNSTABLE in 2h 8m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"da7c5667ddd7bc08cbefe443387ec4f854951d76","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-22 01:32:13.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fd932292d0a942b88516ca4344db90f9 : SUCCESS in 1h 13m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ccd4e6206b7b48a18679bcaa95406801 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":12},{"id":"ba967fd5c631ab5e100dc3ba15eeaab2ae2b640d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-03-23 00:23:36.000000000","message":"Patch Set 12:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/17/913217/12/check/fujitsu-eternus-dx-fc/46ce569 : FAILURE in 2h 31m 07s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/17/913217/12/check/fujitsu-eternus-dx-iscsi/4d51648 : TIMED_OUT in 3h 38m 04s","accounts_in_message":[],"_revision_number":12},{"id":"d47af434b82bfc8b667c204d5353567b75786ec8","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-24 05:27:32.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e6fe3fd0bffb1845c0ece25b79a2182ef40dd1b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 06:00:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d55ebfdfdd740838dc819d6aff29135\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c35527a0af324b8c98c054df9ea8e24e : FAILURE in 15m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05642f6b9c1a4c9db1673b1207eff740 : FAILURE in 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a66d63cea9ac5a31a90eb53c28847f28ec423958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 07:52:00.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaffee7dc9d84bf5881059179bf9cc06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed57fb8a0ea24d57b3caa2b6e0ab8ebf : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbbb9a78c19f425e81c6bddee67c6733 : SUCCESS in 11m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e04118e71034057879d96d4e7d352ef : FAILURE in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffdb662808e47258dc30af213a9f757 : SUCCESS in 16m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/bb83167710974ec8900a860ec4874f79 : SUCCESS in 51m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b52656477c2642c2a37cd0f0fd483446 : SUCCESS in 53m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8303e3fdc9e5496aa47a16a2757c8742 : SUCCESS in 1h 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea24577e83d546238d1d6e3aff839c49 : SUCCESS in 1h 03m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2b86a7cde8045eeb7782ed81fb81ad5 : SUCCESS in 20m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bae812b2b94f4aa8a1722b0d218888b7 : TIMED_OUT in 41m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/01e83c11707d45cb982ffde0019025c4 : SUCCESS in 7m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/310ee99ec6e644169aba9e9e3704adb7 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea49e0a495ad4de89f971fcbd398bf32 : SUCCESS in 6m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a6e6cdfb78ff48a08823e98cafc4960f : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a35b690e166e4eed9b188c55f641e327 : FAILURE in 18m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a1ed2aa93fc648fb8bfcd6ae86aaeab4 : SUCCESS in 7m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b25889589ee04b68ac7ef83d974a7d5b : SUCCESS in 1h 04m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0f03173095e04237860520b33ab4b796 : SUCCESS in 2h 17m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/50c8475ceaf64be2a5032fd8fec5ab79 : SUCCESS in 1h 37m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/810cc2c645da4840a05f6b7a9f358b1b : SUCCESS in 1h 24m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0ecc37f3275040dc90a237217cab0c34 : SUCCESS in 26m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e3f8b4a2df1540f08b8aee3c94b92409 : SUCCESS in 1h 03m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/be6782902d77499a8beddee88537c78b : SUCCESS in 1h 34m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/12feff8961404774834b020d6b7e213b : SUCCESS in 31m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4077d7a1452740fdb21be1251fa2633b : SUCCESS in 47m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/095a7062efce40a9be83188ff5157568 : SUCCESS in 1h 17m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5334dd6de4c44d7980e65449996aea42 : SUCCESS in 1h 16m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcb5783717e74fd287840e8490e00a9c : SUCCESS in 1h 25m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d8a52b6856044bba5fc65b93f6fbcdd : SUCCESS in 29m 20s","accounts_in_message":[],"_revision_number":12},{"id":"33ee77e4e02c6ade704b6c1d093e5da171ff0709","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-24 08:01:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3675733822da46eca9a120257125ac2e : SUCCESS in 1h 13m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2c18f77c991547ddaea7ca840682d840 : SUCCESS in 1h 13m 54s","accounts_in_message":[],"_revision_number":12},{"id":"353c6833b249810cd85a1d4041d735d06340de68","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-25 13:54:10.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/12/20250325181944/NEC-V-ISCSI : SUCCESS in 3h 48m 20s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/12/20250325181944/NEC-V-FC : SUCCESS in 4h 34m 23s","accounts_in_message":[],"_revision_number":12},{"id":"45d891f75bd1177ec32c99f0094e8caf4aa0d45e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-26 04:08:40.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"b872b7626543ab231fdf273de23cddd568c498e9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 04:08:49.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1580/","accounts_in_message":[],"_revision_number":13},{"id":"cc859ed079ba9ea2fc0efa0b96a825ce1ea11354","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 04:08:52.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913217/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/913217/13/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"10c9d7c65ff8751413b909813f91f17b48b8069f","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 04:10:45.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"893a0c149b973d08b60f0c73543de0c987c3efba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 04:44:22.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb306b5fb706406fa1b6a98032fdefc6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b403fd17093a4679a282b36a8f0dafa1 : FAILURE in 32m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/631cb00218a6475e82ad0da0a70da20f : FAILURE in 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1911406a29780d981f5d71fa3e08018453a3bb13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 06:37:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ef091acc3db47df83832724f9d53225\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1af5725baa554c1e9e9bd66e9c14a0aa : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6507c2adcf6f49a6b78b67035352675d : FAILURE in 13m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be12ddd70bfc407e97118a88ffb061b2 : TIMED_OUT in 2h 01m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5dfd7085eb44818803e7762e092554 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/efeaa937607642329b74169659c13057 : SUCCESS in 53m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d490a3785e694e5d8b62f3e6f13246d7 : SUCCESS in 58m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78381267fcf244999027ba86d45c6b35 : SUCCESS in 1h 18m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7eb48eee5330433ea1a000c4c8e5c405 : SUCCESS in 58m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e103baa66ad444b39e849b38abac27b0 : SUCCESS in 18m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a2201fcbbbfd4d8a96a8129adfd8deac : TIMED_OUT in 52m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b8977a7297d2436eb5f31f218192e5ea : SUCCESS in 6m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea860e5df2414cd58d912864e812d1c6 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/400defee22f8480da17f7e6ffc5bf5d7 : SUCCESS in 6m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e35b805877d2442ab66ab0ed6725b348 : SUCCESS in 8m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ef419600074b4cb9a1faf2be762701d3 : FAILURE in 39m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/688f95dd922c4b4ab68324adcbd2cf98 : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e4b01ba46e2b46f090d6bade4687cd4c : FAILURE in 1h 46m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e02265064b3483aa82b0387c694bd4c : SUCCESS in 2h 08m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/59870e7cd574411fbf910eb64d2d337d : SUCCESS in 1h 58m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/62dc0aae95f6482ebf36007455a76b30 : SUCCESS in 1h 33m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/01ecd4347d9b4d5ca87cb7b6d8933ca6 : SUCCESS in 16m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/93caf672367e40159612fd47fa5a69ca : SUCCESS in 1h 05m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6530b86f4eba496db69970d722bb211d : SUCCESS in 1h 32m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/35008fa8e32d48ea81421ec93c5e9ccf : SUCCESS in 41m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4831497db53f43fcba33a0cb0f062bf8 : SUCCESS in 1h 30m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/99e3b29de9d44aa6b5a8ffbf2ed74f76 : SUCCESS in 1h 42m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8ef7b01674cc4f2480d003c014eb1b7f : SUCCESS in 1h 28m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/427d28594bb5459ba502d4bfa7c38f1d : SUCCESS in 1h 30m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0989d808b634a9085d4b9988dcaee84 : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":13},{"id":"79045fe61094d54ca19fd0a6df87a255120ffdad","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-26 07:28:23.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-13/FC/ : UNSTABLE in 33m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-913217-13/iSCSI/ : FAILURE in 3h 4m 7s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"46ce2175787a0915f8ad25f71c69c7cb28e5ccae","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-26 09:28:57.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/74a70692235f401b8261124948503b94\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef46391cbc14433f90c94ac987e93478 : SUCCESS in 46m 26s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e1f00c5f1cce418093b90b5b9a8ca4a1 : SUCCESS in 49m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4a7652770ae14ea2ae1563e6d0364a4c : FAILURE in 1h 57m 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c45e2cfad0934ccebf017ec01c101762 : SUCCESS in 41m 12s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dff27f4f0f4948daaf3d8fcf4fc2ceec : RETRY_LIMIT in 45s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/254bf35835894391bdd121aff0e94097 : NODE_FAILURE Node request 200-0000035692 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a81fcf3887fc458ab1f45fc38de88dea : RETRY_LIMIT in 44s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dc0a64de929740bcb14494d1916d92d4 : RETRY_LIMIT in 44s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"e21b73f3bddf72acaaf6786e03388e59f0b9bdd5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 13:02:55.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913217 : FAILURE in 33m 20s","accounts_in_message":[],"_revision_number":13},{"id":"752fc4f73b42ed3854f2bce9357a0894131fd5c4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 13:22:09.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913217 : FAILURE in 34m 15s","accounts_in_message":[],"_revision_number":13},{"id":"b51929d47043a12ea77b3a640aa015a35d45426a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-26 15:54:30.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"bd2684485aa2cdee2ff4ab3b850590ce8dd84201","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-26 22:56:53.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/583/.","accounts_in_message":[],"_revision_number":13},{"id":"ca9fb40ea80820cd655f8ec94973dbd607535f58","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-27 00:07:33.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7ce7109fab824439affd8ebc770f0b91 : SUCCESS in 1h 12m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d5c6046062b5465c9d112323ed7cd0b2 : FAILURE in 1h 52m 26s","accounts_in_message":[],"_revision_number":13},{"id":"03a9cb72019a724927fdf84f8d8653ac582f0a4e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-31 20:41:04.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/13/20250401014007/NEC-V-ISCSI : SUCCESS in 3h 56m 47s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/13/20250401014007/NEC-V-FC : SUCCESS in 4h 00m 54s","accounts_in_message":[],"_revision_number":13},{"id":"5dc18e82ee93277913eb83216090c49288f5106f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-02 01:31:26.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/13/20250402043705/NEC-ISCSI : SUCCESS in 5h 31m 18s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/913217/13/20250402043705/NEC-FC : SUCCESS in 5h 54m 18s","accounts_in_message":[],"_revision_number":13},{"id":"4a04b11fc7f1009911c39d22e627e87da37d83c3","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-04-08 00:21:48.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"041ce51960022b2f3c2275291b922658a36cd1a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 01:07:10.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0fd6b2829ed4ab789102257b3820b00\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04c6c6cb9d424893b2f925afcd3f1d1f : FAILURE in 42m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3b7cb9f894f241e893359f7c534534d7 : FAILURE in 17m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ef6028873f0d4086beb41b53b62e04a1 : FAILURE in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"68f4bba4f69319d819ef18227fd913e866970bca","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-08 01:46:47.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f3fca823b53547af9b2cf2b0f21b7f72 : SUCCESS in 1h 15m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/81ec379c27694cfd8b55baa05ac46dbd : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":13},{"id":"c76f91d598540b7001787084f826c32017b7451b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 02:54:58.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd9ae02899cc4824b52f55f72d876b6c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70fb23cb56394a849f638b38f3b6fa81 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ab1ba15a04d4bbebb34dad9bdeb7eef : SUCCESS in 7m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c32cc346750542ae8548302da5c0d5a6 : FAILURE in 10m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77eb2905a6654a7a99e2dfe036111d77 : FAILURE in 13m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca06dd8270a34e85882cdaec34279139 : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/71626ecd53b04a18810c77fb38b87464 : SUCCESS in 53m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/122a69b988f44683b5e3b5870846007a : SUCCESS in 58m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52f3daf310aa4b8784cf89d3568b5856 : SUCCESS in 1h 28m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0471fdb4afe740c688dbd1296daa09c7 : SUCCESS in 56m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60ff4bd009ad47dba16a985551620574 : SUCCESS in 18m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84f95bff423244e2ac40a0c55659997f : TIMED_OUT in 41m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/96507bb77dd94434a2e896b5f7425ac8 : SUCCESS in 7m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ed212d38c52465b90a59ad5e7050ac4 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2af878a8a2e24b8aa3c35c6845559e71 : SUCCESS in 7m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3067cc71aeee4942bead98acd11b836d : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/70ce527d38fe47a1aec7a1456a9389b6 : FAILURE in 34m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b0338bf43e04497b19dbeb9cba05aa9 : SUCCESS in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c6cee60e691246f0a4932a04e4b1a23a : SUCCESS in 1h 39m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/45ce0d8c4b6e45b3b6111b6443a070b6 : SUCCESS in 2h 27m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e53f2fceeb96430b957a8daa756f21f9 : SUCCESS in 1h 18m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4d593b730cb5487f9fa4560c507b9a5e : SUCCESS in 1h 27m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a90320c4cd314576b254274c86f0cb2a : SUCCESS in 25m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/69cbdf3db71b447784c0e4fd4835a72a : SUCCESS in 1h 03m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ef4bb352b64042deb12c86b9fa87e93e : SUCCESS in 1h 56m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/faff0c31179e4731b08867bca88f99fe : SUCCESS in 31m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/525f5d068e9e4c0f81f048bee23d0211 : SUCCESS in 1h 20m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f170fbfbc43a4224bb339d64d464548a : SUCCESS in 1h 38m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d18216f89904e26a07a2ed6f19fe794 : SUCCESS in 1h 31m 15s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e86885f9cc8452c8969df2bca2e7420 : SUCCESS in 1h 16m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83180b5e4607439f88fce78221cbfb23 : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":13},{"id":"58932f700f432ee9c4f1fc7c8b97773c4507c4e1","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-04-11 05:00:22.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d13129feb53fbbe243e79aca56242d6fa428917d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 05:38:29.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24845d74de6498084ed7043258e0ba8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9aa33a7ade01429b9b9f684b988e12a3 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/492a8d9892f44ece92da649f64db2019 : FAILURE in 22m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/709e311cd8b940b4af3db1127091fa4a : FAILURE in 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"30963ac607030254b1096a11822a249e1db6e185","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-11 06:21:46.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ce77ada39dbc4c79a39fc28ae76b967a : SUCCESS in 1h 10m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eec22f46dac04a21bbd23ba5eb224207 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":13},{"id":"d9f6488b638687a0dd53557a2b2846087cf713db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 07:10:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/868a13115af541bfae463d6367086f92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91b8596b42547d8b49b1e9c5f989b0e : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/573e975b66bf4b7b9293735dcee6992d : SUCCESS in 10m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed70464759af4a0c857d539b645286d4 : SUCCESS in 10m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e65fd6164b9d46b9bd87fd568e4e4454 : FAILURE in 15m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b01851d9f7d4f918ed9143cd994a779 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/3edf62cc234d493fa3fb5632844cb598 : SUCCESS in 39m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/794945de3c9c4fd2b221ac5d5e127f14 : SUCCESS in 43m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd83de8b19a74c7ea75fc27c35d28514 : SUCCESS in 1h 24m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/506e9ed6a51c42119e0ab069d5ddb7fa : SUCCESS in 58m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dc15d7fa2b54ea3a4174a577930eb49 : SUCCESS in 15m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3dace9c4c09d42fa8d08ab7bd6e75828 : FAILURE in 40m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/15b2674cffb5406e8cd4922be4a5383f : SUCCESS in 6m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ccbeccd52ae44e18b5070f3b3b06c422 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22d9b591d4484125a9d30e1e1b90350f : SUCCESS in 6m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f887a2294ed3438da412b8b915c4e500 : SUCCESS in 8m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6c6f739834474d2f93653ba0ae657e86 : FAILURE in 26m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19650931f112465d8b1d2324683460fc : SUCCESS in 7m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ce382e79b52460d98652b255fff56a1 : SUCCESS in 1h 52m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c5cdb1330f31477e82adef01206ab625 : FAILURE in 1h 25m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1e62359de6224f73991c39e761591ce2 : SUCCESS in 1h 46m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1d2ae3d0edc24f13b0c53263af540d84 : SUCCESS in 2h 01m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c831686613e54fec87f6eff644b74575 : SUCCESS in 33m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bee9023d526841a483fb6396e3f047bd : SUCCESS in 1h 06m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/43e1203ebfb8468889370f0cba1b0c2b : SUCCESS in 1h 26m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/80f9bc2d8c3d492d9ddcc882011732b8 : SUCCESS in 26m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8c99525041d24abdb9de6a8753dfc3cd : SUCCESS in 1h 09m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ecc4a616d1e744878a4fb379c0d5806b : SUCCESS in 1h 08m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0fb2fee0807d48bbae517424dccf77d3 : SUCCESS in 1h 39m 11s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78eedc4338fb49c1b55cb00f1aaaa2e6 : SUCCESS in 1h 17m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee25bc219426449c9923ee7ace95f788 : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":13},{"id":"240ddd0d1478fccc44c5555dfdd4ed3f71c45844","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-11 10:27:34.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"a4b8b0ede6d551a56af0f7a481e12b0c479e7f9d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-04-11 13:00:57.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/913217-13/Seagate-iSCSI/ : SUCCESS in 1:56:39\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/913217-13/Seagate-FC/ : SUCCESS in 2:07:09","accounts_in_message":[],"_revision_number":13},{"id":"5497bd499ac36abd94d5c724e546d87097dead03","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-04-11 17:26:51.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/913217-13/PVME-iSCSI/ : SUCCESS in 1:57:45\n- PVME-FC https://os-logs.seagate.com/dell-ci/913217-13/PVME-FC/ : SUCCESS in 2:06:50","accounts_in_message":[],"_revision_number":13},{"id":"1feaac50b8fdea2adaa273f73bdae8ba6c6f79ba","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-11 21:55:20.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/913217-13/Lenovo-iSCSI/ : SUCCESS in 1:55:06\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/913217-13/Lenovo-FC/ : SUCCESS in 2:06:44","accounts_in_message":[],"_revision_number":13},{"id":"1662df2a4258118e0da65dcb557dd42beee68d40","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-10-31 18:02:39.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0e205687f17b33eb5a30057ee3ccbc497f93a0b8","author":{"_account_id":38554,"name":"Lucas Lafontaine","display_name":"Lucas","email":"l.lafontaine@jp.fujitsu.com","username":"ftilucas"},"date":"2026-02-02 08:44:17.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"d75445c68a5f510aa5487bbdd942e680a3f5a32c","revisions":{"84dda08369fc9c8103df8158d5684fe2808c822e":{"kind":"REWORK","_number":1,"created":"2024-03-26 01:42:47.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/1"}}},"commit":{"parents":[{"commit":"38b70a2df9ea872da6431fcec2e3737d68c25b24","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38b70a2df9ea872da6431fcec2e3737d68c25b24"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84dda08369fc9c8103df8158d5684fe2808c822e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84dda08369fc9c8103df8158d5684fe2808c822e"}]},"branch":"refs/heads/master"},"d685e433805d94296dd7c4c563a5da2fd52b13a8":{"kind":"REWORK","_number":2,"created":"2024-04-19 09:37:35.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/2"}}},"commit":{"parents":[{"commit":"ecdcf7b0daaa21e3e81bf42fe38bef76078edd2d","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ecdcf7b0daaa21e3e81bf42fe38bef76078edd2d"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-04-19 09:37:15.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d685e433805d94296dd7c4c563a5da2fd52b13a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d685e433805d94296dd7c4c563a5da2fd52b13a8"}]},"branch":"refs/heads/master"},"c27b21891b44c0ac1713972e66ce3c22a4c069ee":{"kind":"REWORK","_number":3,"created":"2024-05-21 06:14:17.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/3"}}},"commit":{"parents":[{"commit":"8c5130114359b8e6676d375ffb13e86210b8cafc","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c5130114359b8e6676d375ffb13e86210b8cafc"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-21 06:12:19.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27b21891b44c0ac1713972e66ce3c22a4c069ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27b21891b44c0ac1713972e66ce3c22a4c069ee"}]},"branch":"refs/heads/master"},"0135a1f471be2dc90113fdd40e02e2001bcd70bf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-24 10:22:52.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/4"}}},"commit":{"parents":[{"commit":"0be3ed40e581ba73d8f13b6cfbf5ecc6b758b029","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0be3ed40e581ba73d8f13b6cfbf5ecc6b758b029"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-24 10:22:12.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0135a1f471be2dc90113fdd40e02e2001bcd70bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0135a1f471be2dc90113fdd40e02e2001bcd70bf"}]},"branch":"refs/heads/master"},"ecd2bad5b5de25e8306fbf85e0cb4e1b435b3e5e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-27 10:04:57.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/5"}}},"commit":{"parents":[{"commit":"a27998fbfbb1d628a1c425238b2c9b6c0c986370","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a27998fbfbb1d628a1c425238b2c9b6c0c986370"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-27 10:04:30.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ecd2bad5b5de25e8306fbf85e0cb4e1b435b3e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ecd2bad5b5de25e8306fbf85e0cb4e1b435b3e5e"}]},"branch":"refs/heads/master"},"0b24d1e367107071b96c79af7ea7eef51687faa5":{"kind":"REWORK","_number":6,"created":"2024-06-14 07:23:24.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/6"}}},"commit":{"parents":[{"commit":"063c758b7dde9948d435da709ca719ef78ef8e07","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/063c758b7dde9948d435da709ca719ef78ef8e07"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-06-14 07:22:44.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b24d1e367107071b96c79af7ea7eef51687faa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b24d1e367107071b96c79af7ea7eef51687faa5"}]},"branch":"refs/heads/master"},"b0bda07b96f04ee5b3d713f6ac5c44d453aad0f7":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-07-12 07:29:15.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/7"}}},"commit":{"parents":[{"commit":"44ce5c561e6af8a10810f821e5bd2d9b7d73ab08","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44ce5c561e6af8a10810f821e5bd2d9b7d73ab08"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-07-12 07:27:47.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b0bda07b96f04ee5b3d713f6ac5c44d453aad0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b0bda07b96f04ee5b3d713f6ac5c44d453aad0f7"}]},"branch":"refs/heads/master"},"bd5339ebf6cd94a6fc6c4eb0f0f8997bff3c0c09":{"kind":"NO_CHANGE","_number":8,"created":"2024-07-12 09:51:27.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/8"}}},"commit":{"parents":[{"commit":"b27a4542724576b5c78a9809c02de30c88ee8b16","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b27a4542724576b5c78a9809c02de30c88ee8b16"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-07-12 09:50:54.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd5339ebf6cd94a6fc6c4eb0f0f8997bff3c0c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd5339ebf6cd94a6fc6c4eb0f0f8997bff3c0c09"}]},"branch":"refs/heads/master"},"b6abef9ab846809506cd1bb1eeca42d0ac1db9b0":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-08-07 07:43:37.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/9"}}},"commit":{"parents":[{"commit":"e650f79a3d1927b5d5972eaa7014c13182572983","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e650f79a3d1927b5d5972eaa7014c13182572983"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-08-07 07:38:18.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6abef9ab846809506cd1bb1eeca42d0ac1db9b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6abef9ab846809506cd1bb1eeca42d0ac1db9b0"}]},"branch":"refs/heads/master"},"c65ada41cf13652cf50c1754e06e79446d7db110":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-09-04 03:40:18.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/10"}}},"commit":{"parents":[{"commit":"9ae1e918de90719ac97e614d5692bb4682bc0d06","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ae1e918de90719ac97e614d5692bb4682bc0d06"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-09-04 03:20:39.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c65ada41cf13652cf50c1754e06e79446d7db110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c65ada41cf13652cf50c1754e06e79446d7db110"}]},"branch":"refs/heads/master"},"a079ed4278f1785e7d35537e337b1ef09fc29bd3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-01-07 06:27:21.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/17/913217/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/11"}}},"commit":{"parents":[{"commit":"c531dcbd6c0d3eb8fc69526352b7429b31cc22fa","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c531dcbd6c0d3eb8fc69526352b7429b31cc22fa"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2025-01-07 06:26:41.000000000","tz":480},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a079ed4278f1785e7d35537e337b1ef09fc29bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a079ed4278f1785e7d35537e337b1ef09fc29bd3"}]},"branch":"refs/heads/master"},"1e631afc64d4accdde5be4181e7ed79008061643":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-03-21 07:53:29.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/17/913217/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/12"}}},"commit":{"parents":[{"commit":"2c11c345438425a418930700c330779c7d32b265","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c11c345438425a418930700c330779c7d32b265"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-21 07:51:54.000000000","tz":-240},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e631afc64d4accdde5be4181e7ed79008061643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e631afc64d4accdde5be4181e7ed79008061643"}]},"branch":"refs/heads/master"},"d75445c68a5f510aa5487bbdd942e680a3f5a32c":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-03-26 04:08:40.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/17/913217/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/913217/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/913217/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/913217/13"}}},"commit":{"parents":[{"commit":"874d8fe30179edf3744a6e6fe47ef80bb7c54c1f","subject":"Fujitsu Driver: Support get manageable resources.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/874d8fe30179edf3744a6e6fe47ef80bb7c54c1f"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-26 01:42:24.000000000","tz":-240},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-26 04:06:32.000000000","tz":-240},"subject":"Fujitsu Driver: Support manage volume and snapshot","message":"Fujitsu Driver: Support manage volume and snapshot\n\nThis patch supports manage volume and snapshot on Fujitsu Driver.\n\nImplements: blueprint fujitsu-add-manage-volume-and-snapshot-support\nChange-Id: I9168dd609a42b0f60dd210f1184fb50065f409ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d75445c68a5f510aa5487bbdd942e680a3f5a32c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d75445c68a5f510aa5487bbdd942e680a3f5a32c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
