)]}'
{"id":"openstack%2Fcinder~913436","triplet_id":"openstack%2Fcinder~master~Id6a1452c2c197a58677bf181470f54565fbd263b","project":"openstack/cinder","branch":"master","topic":"bug/2055517","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2024-11-20 09:33:01.000000000","reason":"Change was marked work in progress"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2024-11-20 09:33:01.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Id6a1452c2c197a58677bf181470f54565fbd263b","subject":"Allow clone encrypted image to encrypted volume","status":"NEW","created":"2024-03-15 17:37:20.000000000","updated":"2024-11-20 09:33:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"28f416887513415d42432674c96771f8f1742c19","_number":913436,"virtual_id_number":913436,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-16 21:49:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 17:37:29.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 17:44:19.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2024-03-15 17:46:49.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-15 17:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 18:38:36.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 19:51:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-15 20:49:44.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-16 00:36:48.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-16 01:10:45.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-16 01:21:41.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-16 05:41:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-16 18:34:18.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-16 23:44:16.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-17 02:18:08.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-03-17 08:07:32.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-03-18 07:11:02.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-20 08:17:15.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 14:09:54.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-21 05:00:12.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-30 05:36:59.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"bdd1ab4c60251be0814a005abc2e0354102ce9fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-15 17:37:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11adf5cffd5cb093a86f015072faad9317fed03a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 17:37:29.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16651/","accounts_in_message":[],"_revision_number":1},{"id":"78d74a7344ff759bdd086e6c69ca05f189e9b9a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-15 17:37:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16651/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"660a0587a02a9ee74dcaef969d9ec75c725844cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-15 17:46:49.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a3f828960e3b432ca0222873518b1e26\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8489c184a6f5487d95251050ff8593fa : NODE_FAILURE Node request 200-0000002796 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"3721e380229f97b47c468216fdb6b8e48318c4e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 17:51:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48abac14a6fa4e269807070d669209e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b89930a31d9453b82101efd6b994ef9 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1d5436a4db9465a90a8fa46c639bca1 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ff97108ccbb45aba6844237e04d7f14 : SUCCESS in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52d122d72db83c9e0ad4e6e047a884fb051cafaf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-15 18:38:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913436 : FAILURE in 59m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d88c8694e73782bc0b0462d394e17c5a85670943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 19:51:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/21c8a92ee53d48f0901e73d9e32c431a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f686fb79f99495ba3c949591940dcad : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28476762ef2c42189541893f85238d79 : SUCCESS in 10m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52e6d3c15ac74c07a2262a74ff6a28de : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4000180f9534fab8358d01f3e24575d : SUCCESS in 13m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/37100dcad388430fa4618906f38693e5 : SUCCESS in 1h 04m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/40c44a041f8643e8bf68337797b8b2df : SUCCESS in 56m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9915519403ab479c80acb7a2232c884e : FAILURE in 1h 00m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b21a2caef5a4e4ea87053cd2abc9365 : SUCCESS in 55m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dbf43c82aaa4217b29573b44f9963c1 : SUCCESS in 11m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65e27b9bd75e43d3891011b962394f9e : SUCCESS in 15m 43s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ed47512531f64db7b4ae739d6fc3f727 : SUCCESS in 13m 01s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d83066918420437d9f8f0d891c408bdc : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4cc3c930cd16407ebb5f9081bc961e3f : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/93112beeb50e42cc875b057357e0f6f9 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7f0c20ade36e43d69a1c482795f4e289 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/898510d99c2147b8b64528a54475c389 : SUCCESS in 36m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2bda4b7f10b74be8840d138d6e73a388 : SUCCESS in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c0f73cc194184d81ada1e2a5c782370f : SUCCESS in 1h 57m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cbf0db0dd34344eaa326bf3b1a8bc87b : FAILURE in 2h 11m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3864fdb214ff40e6ac8ffb4edce8378c : SUCCESS in 1h 19m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cfdaf935005e490a95a82decd0510dc4 : SUCCESS in 1h 39m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bfb6f42150ed463f9daee90c398d6c77 : SUCCESS in 37m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f2f1d605ea3149f3850a0a834e7442f6 : SUCCESS in 1h 10m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/63c0cc79315f4d2d99527367b85f05e4 : FAILURE in 1h 23m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/24b37aef8bf1416d860c23cc6d6b65c3 : SUCCESS in 45m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3c55e9b414c04cf9b4b463d7af1f1863 : SUCCESS in 1h 05m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0b9759a445be47a18079605032f7bd2a : SUCCESS in 1h 40m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b3bb992f2e0471ea3fd8e4eadaddee0 : SUCCESS in 1h 20m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b465fbe2d1949b398d91afdc6f5aa40 : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2cc19f06971f573aa26c1639b6fcbfc76dc0c3e5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-15 20:49:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-36-913436-1/FC/ : SUCCESS in 2h 23m 36s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-36-913436-1/iSCSI/ : SUCCESS in 3h 12m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"794a8f765b49d2bd846a289bfd6787c2067c8234","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-16 00:36:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/63/.","accounts_in_message":[],"_revision_number":1},{"id":"b5b6d8f7973dc8b16eb70fc1caf3ff3cb40c099a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-16 01:10:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/36/913436/1/check/fujitsu-eternus-dx-iscsi/1786463 : FAILURE in 3h 35m 05s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/36/913436/1/check/fujitsu-eternus-dx-fc/e37a478 : FAILURE in 3h 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d0e1d74dc53b492f279c2f024d6e14f3ebe4cee7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-16 01:21:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913436 : FAILURE in 33m 39s","accounts_in_message":[],"_revision_number":1},{"id":"116114c0a2ad1c6c754332792f75b912fb16357d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-16 03:44:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913436 : FAILURE in 1h 57m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bd499460059c640f1429ee9068b1f6209732cacb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-16 05:41:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_Unity_FC/f3bfcf1/DellEMC_Unity_FC/1521 : SUCCESS in 1h 15m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_Unity_iSCSI/0d1086e/DellEMC_Unity_iSCSI/1553 : SUCCESS in 1h 38m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"db814d2276c714eda5e119071c7c47fabd8b8f99","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-16 18:34:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/36/913436/1/ : RETRY_LIMIT in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c994441f2adf503368acee1515afd858964af11e","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-16 19:24:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d25a36d0d2649c161c9b24b5a06da50a8a0d064c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 19:42:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e2c959a719d4891bde1e32c4f9797c8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f90fa653e6a248928648b776a4a4e88d : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b5caac3139a448c96cb0e7f385f2ac4 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75c97890405b4cc8a84c2074ac576fde : SUCCESS in 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a44612cb6a3797ee28dd84a775ae5d7d78b79ca5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-16 19:46:43.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c80b5094fea049d09564e3651645ae8d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5e4ce63acdf14ecbb3d90a7a16c6b20a : NODE_FAILURE Node request 200-0000002799 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"100ad4f77e06f44b33a67dfc3c3e74e61fcff2e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 21:49:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd8ca16ffc844f78bb2bf9b715aeaa81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8011d6914b1747b5ba96ae74ebcbe93b : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9268ecc82ae4399a602297db14905a8 : SUCCESS in 11m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/818feb24a4c3401d9489e7308bc28366 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe1f886e44f4bd9bba46857d4999638 : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/461fff8a8478499b8f0f62554214ea66 : SUCCESS in 49m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/39193e8f0f4940a5bfb8285d30ea2a4a : SUCCESS in 1h 05m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/caaee4f8b0144635bb0e4c0f18c77023 : SUCCESS in 1h 21m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/246ab73d3c03414dafe8550164aa937b : SUCCESS in 50m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57ae9f082a1f4892a1aa98bfb7ee23b8 : SUCCESS in 12m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/80f34af2a7e04893bc0f4a072b57388a : SUCCESS in 12m 29s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/57dabaf353514b78a1960c362a32e3dc : SUCCESS in 13m 21s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c6a7ac54946847beac4e76e38112dee5 : SUCCESS in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/43ae90ed706547838e54847b7fd39525 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/944162e716e243358cf193bc42fa2a05 : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d420d785cdc740d8878a2b3eeae0e8ef : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c545074a679746188064b92029bf9f3d : SUCCESS in 44m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/61e5861bd1f74f07bfd82442575fc8e2 : SUCCESS in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/635baf7cd1ac4320a527e10bcfc09118 : SUCCESS in 1h 34m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/45c2023480634263b0e57280d5efa4cf : FAILURE in 2h 15m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/76677a6e4ebd4759a2c684bd13ed0fbc : SUCCESS in 1h 20m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a4fe60fd45514d0abce17e298e02862b : SUCCESS in 1h 33m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a8dafbdf800e4593a3d748d78e65afbb : SUCCESS in 23m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/96f5405def6b4d15ba2ff8aa0b0f586c : SUCCESS in 1h 03m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6de68de803c34880b0182cba4f7d9f54 : SUCCESS in 1h 29m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6d3bc8c81b614d6099a8fd89a3d58d7e : SUCCESS in 45m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cd109687f7a5464885fcf200e750de15 : SUCCESS in 57m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f36a8099c168427ab13661f978086b36 : SUCCESS in 1h 52m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1aedd6a9e5834bbd825e8e33f713ef38 : SUCCESS in 1h 28m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce91bdb52fa443a4a825803edcb6bace : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":1},{"id":"225e60286347d2846519f903eb3af63c574214a5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-16 23:44:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/913436/1/upstream-check/cinder-tempest-plugin-ontap-fcp/aea2a5f/ : SUCCESS in 1h 33m 06s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/913436/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/14bad79/ : SUCCESS in 43m 35s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/913436/1/upstream-check/cinder-tempest-plugin-ontap-nfs/d790670/ : SUCCESS in 1h 21m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/913436/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c70c5b7/ : SUCCESS in 1h 20m 29s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"832a9b9c9bb2a92ea7b55570a0948f0c05075cd3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-17 01:00:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_Unity_FC/031f8e5/DellEMC_Unity_FC/1525 : FAILURE in 1h 27m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_Unity_iSCSI/a5e69ff/DellEMC_Unity_iSCSI/1557 : SUCCESS in 1h 44m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"11508d6426962bc897328d468f85670729f7a190","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-17 02:18:08.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_PowerStore_iSCSI/e3259d9/DellEMC_PowerStore_iSCSI/1169 : SUCCESS in 1h 16m 13s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_PowerStore_FC/b61be27/DellEMC_PowerStore_FC/1339 : SUCCESS in 1h 28m 33s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_PowerStore_NFS/04d1627/DellEMC_PowerStore_NFS/857 : SUCCESS in 1h 06m 57s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_PowerStore_NVMeTCP/4b6d7ad/DellEMC_PowerStore_NVMeTCP/719 : SUCCESS in 1h 43m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"e5f4bf905882814648c6b6313f49dfba2852a34d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-17 08:07:32.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/17_03_2024_08_02_27\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"4d97f3205fb33a9abecf6701c8c6ab1b8d5e3ba4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-17 10:13:10.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/17_03_2024_10_08_12\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"2d8ed21916c5775e80cca765c328f99a20c64a20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-17 21:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_03_2024_21_48_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"6afc0915f23adbccc2511b3af19c4a2449b4ebfa","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-18 07:11:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/913436/1/20240318123624/NEC-V-ISCSI : SUCCESS in 3h 14m 37s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/913436/1/20240318123624/NEC-V-FC : SUCCESS in 3h 34m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2de34fd4b59f2ce4e2cbefa46c6ff54b60cefc55","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-20 08:17:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_PowerMAX_FC/624561c/DellEMC_PowerMAX_FC/352 : ABORTED in 10h 30m 01s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/36/913436/1/check/DellEMC_PowerMAX_iSCSI/c86f7db/DellEMC_PowerMAX_iSCSI/372 : ABORTED in 10h 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"012e81f51e810a83891db75dd0b7845e94e673f8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-20 14:09:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/913436/1/20240320143042/NEC-ISCSI : SUCCESS in 4h 11m 23s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/913436/1/20240320143042/NEC-FC : SUCCESS in 8h 39m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c288466179a7b78796cb804b7319f1fe62a38d8b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-21 05:00:12.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/36/913436/1/fc : SUCCESS in 0h 39m 58s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/36/913436/1/iscsi : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":1},{"id":"eff03dc25398c386c69a035117a1d19e5fd4a14b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-30 05:36:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/913436-1/Lenovo-iSCSI/ : SUCCESS in 2:07:56\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/913436-1/Lenovo-FC/ : FAILURE in 3:03:48","accounts_in_message":[],"_revision_number":1},{"id":"28f416887513415d42432674c96771f8f1742c19","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-11-20 09:33:01.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23915c1854451618276c63c5bcb2f03a970d36bf","revisions":{"23915c1854451618276c63c5bcb2f03a970d36bf":{"kind":"REWORK","_number":1,"created":"2024-03-15 17:37:20.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/36/913436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/913436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/913436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/913436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/913436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/913436/1"}}},"commit":{"parents":[{"commit":"942289ecb3786ed383b5106ae97918fe815658ba","subject":"Merge \"Deprecate Windows OS support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/942289ecb3786ed383b5106ae97918fe815658ba"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-03-15 16:35:12.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-03-15 17:09:12.000000000","tz":480},"subject":"Allow clone encrypted image to encrypted volume","message":"Allow clone encrypted image to encrypted volume\n\nExactly like what we did in copy-and-import image when create encrypted\nvolume from encrypted image. If the image is encrypted, we will copy\n`cinder_encryption_key_id` from image metadata to volume. That means we\nshould be safe to try directly clone from encrypted image.\n\nRelated-Bug: #2055517\nChange-Id: Id6a1452c2c197a58677bf181470f54565fbd263b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23915c1854451618276c63c5bcb2f03a970d36bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23915c1854451618276c63c5bcb2f03a970d36bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
