)]}'
{"id":"openstack%2Fcinder~913816","triplet_id":"openstack%2Fcinder~stable%2F2024.1~I1f025542a2509e36919ece01b29064377dbbe189","project":"openstack/cinder","branch":"stable/2024.1","topic":"I1f025542a2509e36919ece01b29064377dbbe189-stable/2024.1","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-03-26 04:10:02.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I1f025542a2509e36919ece01b29064377dbbe189","subject":"HPE 3par: Unable to create clone of replicated vol","status":"MERGED","created":"2024-03-22 12:17:45.000000000","updated":"2024-03-27 05:12:29.000000000","submitted":"2024-03-26 06:02:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913816-I1f025542a2509e36919ece01b29064377dbbe189-stable/2024.1","cherry_pick_of_change":884594,"cherry_pick_of_patch_set":13,"meta_rev_id":"fa5193f3f62090da97ecbf66f24f3e8e296204a9","_number":913816,"virtual_id_number":913816,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36345,"name":"saikrishna","display_name":"saikrishna","email":"gksaikrishna@gmail.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 06:02:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-03-22 13:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-03-24 17:27:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36345,"name":"saikrishna","display_name":"saikrishna","email":"gksaikrishna@gmail.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-26 03:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36345,"name":"saikrishna","display_name":"saikrishna","email":"gksaikrishna@gmail.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-26 03:59:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-03-22 13:30:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36345,"name":"saikrishna","display_name":"saikrishna","email":"gksaikrishna@gmail.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36345,"name":"saikrishna","display_name":"saikrishna","email":"gksaikrishna@gmail.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-22 12:24:12.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-22 12:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-22 12:54:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-03-22 13:06:19.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 13:07:52.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-22 13:09:18.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-22 13:30:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-03-22 13:40:16.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 13:43:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-24 17:27:35.000000000","updated_by":{"_account_id":36345,"name":"saikrishna","display_name":"saikrishna","email":"gksaikrishna@gmail.com"},"reviewer":{"_account_id":36345,"name":"saikrishna","display_name":"saikrishna","email":"gksaikrishna@gmail.com"},"state":"REVIEWER"},{"updated":"2024-03-26 03:59:53.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"0a69f1f7acdc60504be903c689508e09c86025a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-03-22 12:17:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1438efd9e5fdcfa7946364312d7d8c555af53d28","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-22 12:24:12.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ba79e4fc8f1346ceac42ad11bcca9fb3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d3cd1c1bcc13447fa097a1d4c7e1a37f : NODE_FAILURE Node request 200-0000002868 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ee0f0a3429f9be09fd73a410a5fb6a937c347064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 12:47:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e1fbd3d86b748cd8dffac2936c5e9db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51b52962f9246ed88b7274d1e4893fc : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac697137476e4279a3383e1570a437f1 : SUCCESS in 12m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c393a16f30046eaa32ff13334dd3c5f : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30ddfb2002d4571aae5fc0c00c40d99 : SUCCESS in 15m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbd6aea395f6478eacd02fa506fb14a5 : SUCCESS in 11m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2b2f25a46cd34688a997e0574c45b6fd : SUCCESS in 13m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d43adc99801243b1a6386bccd28d49fb : SUCCESS in 6m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5b2357363a54fc1ba648d2b3172df3c : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb574c9b0abe41e1bdb650aa0b8b2d7a : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c18f4bed96548a19f677a234a8454c1 : SUCCESS in 6m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db3cfdadbcae48f89f8d59070aa5bd1e : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cbf0f6d35973aa1ce1d291d433ab80b6e260198","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-22 12:54:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_03_2024_12_48_00\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"5801477373b08197a806a73c90d4f0d585013f86","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-22 13:06:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/913816 : FAILURE in 47m 04s","accounts_in_message":[],"_revision_number":1},{"id":"49c6ca010f51f3f5527fc224f4bf25a61d6ffdfa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-22 13:07:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/913816/1/upstream-check/cinder-tempest-plugin-ontap-fcp/0974237/ : RETRY_LIMIT in 3m 58s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/913816/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/13e9f8c/ : RETRY_LIMIT in 3m 25s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/913816/1/upstream-check/cinder-tempest-plugin-ontap-nfs/3e2a2a2/ : RETRY_LIMIT in 4m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"79540e4215f4d350f40cd13ee19a5c04344b2081","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-22 13:09:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913816 : FAILURE in 49m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c6f810b3b663a44ab3f97f55cda8d4062c14ebde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 13:21:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fe6304e062f4c7e8e54ba6b2a497525\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca50401392dd4fcb8f05adc7fbf49aca : TIMED_OUT in 1h 01m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e7cf9cb644614ed6bdc964cae306f41d : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b3910f059ed54982b82903de4a1da869 : SUCCESS in 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b18518261cce8bb1d622278db5a475d05d42649","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-22 13:30:35.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c92335a91659936d1c7987cf6559e52126338f8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-22 13:40:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/16/913816/1/ : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":1},{"id":"88ee076059416371e3da2c06c13ec0c5a4667b62","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-22 13:42:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/913816/1/upstream-check/cinder-tempest-plugin-ontap-fcp/92ecb5e/ : RETRY_LIMIT in 3m 45s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/913816/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/9118fbb/ : RETRY_LIMIT in 3m 48s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/913816/1/upstream-check/cinder-tempest-plugin-ontap-nfs/dba1648/ : RETRY_LIMIT in 3m 42s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"518b3b148170cc4e582990c73503585b35b074a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-22 13:43:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/122/.","accounts_in_message":[],"_revision_number":1},{"id":"820b1d635b6360d4d74b06dcb079f946838a53d2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-22 14:03:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/913816 : FAILURE in 50m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9417c5852f9b3c9fb27a2cadc48cad6ee7649a8d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-03-23 09:48:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90457245cbf19305601308fffcb87ac06ce8fcaf","author":{"_account_id":36345,"name":"saikrishna","display_name":"saikrishna","email":"gksaikrishna@gmail.com"},"date":"2024-03-24 17:27:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef58dc2964ad79c543ea7b1810da67b67c95829e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-26 03:59:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f27b0e22a1a68b8c31ffd29c4c007c2efe7cd1f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 04:00:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c014e0a20c70a7e2594c56d9d86dc73ca07d31a9","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-03-26 04:10:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df7dda7bd15049513c55578042be2f804af5ca6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 06:02:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2337addc210442a49ea08f24faf8c5d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dffec5263ab94006b73745dd68323723 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2cf4addb6e045c8937ec9577363dc55 : SUCCESS in 10m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11be9f6681de48b2aa2087206409f000 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8354f6f3d7964252b5e9cab9026ec1ef : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/fe36238414314bb1ad6cb01fc146f0fc : SUCCESS in 1h 03m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a0230feaaec4ab89ca3388fdc121503 : SUCCESS in 1h 00m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91e25d977de243d8a52eccf5201a8be3 : SUCCESS in 1h 03m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ca6ef222ee14dd99f8509625d39ee57 : SUCCESS in 14m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3dc0ff6d2c6944fb8e77fbd6cb76615b : SUCCESS in 1h 00m 49s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/599c66fd654c4589b8a8835a279861ef : SUCCESS in 1h 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed2c4ed6bc1c4b8580572e0e79c36818 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c99cafbe1a9230da9feca2b268d85950276aae43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 06:02:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"36ad19fbf9ce1859cdd766089329f4323508d5f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 06:04:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c34a70e7014cf4acc84c4a56c6f26c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97bc8de76f6460492e6616c4274fa62 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b94be6f108a418389f27317126cfe81 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"fa5193f3f62090da97ecbf66f24f3e8e296204a9","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-03-27 05:12:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2360e80cd9a75da1eec91b9ee557451a4a7bb324","revisions":{"2360e80cd9a75da1eec91b9ee557451a4a7bb324":{"kind":"REWORK","_number":1,"created":"2024-03-22 12:17:45.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/16/913816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/913816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/913816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/913816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/913816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/913816/1"}}},"commit":{"parents":[{"commit":"7405b8d07a50714e0af96f307de70916d51c87a6","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7405b8d07a50714e0af96f307de70916d51c87a6"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-05-30 19:45:50.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-03-22 12:17:45.000000000","tz":0},"subject":"HPE 3par: Unable to create clone of replicated vol","message":"HPE 3par: Unable to create clone of replicated vol\n\nTwo possibilities of clone volume:\n1] same size, online copy\n\nExisting behaviour: start clone \u0026 return from function.\nError occur because clone is not yet complete and\ncode tries to create vol on secondary array.\n\n2] size is different, offline copy\n\nExisting behaviour: (i) create new replicated vol.\n(ii) during clone operation below error occur:\nVolume is involved in remote copy\n(iii) Since clone operation fails, delete new replicated vol (as cleanup).\n\nTo overcome both possibilities, code changes are done.\n\nFor clone of replicated vol, create offline copy only.\nSteps:\n(i) Create new vol without replication.\n(ii) Perform clone operation; wait till completion (offline copy).\n(iii) Create vol on secondary array.\n\nCloses-Bug: #2021941\nChange-Id: I1f025542a2509e36919ece01b29064377dbbe189\n(cherry picked from commit 577e2ee108a5e3ec2b727928ee4e9c8b35331339)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2360e80cd9a75da1eec91b9ee557451a4a7bb324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2360e80cd9a75da1eec91b9ee557451a4a7bb324"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
