)]}'
{"id":"openstack%2Fcinder~914230","triplet_id":"openstack%2Fcinder~master~I940049a9d58252980a8be47bdc93924e91436238","project":"openstack/cinder","branch":"master","topic":"fujitsu-driver-update","attention_set":{"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2025-03-21 11:47:15.000000000","reason":"Someone else replied on the change"},"37834":{"account":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"last_update":"2025-03-21 11:47:15.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I940049a9d58252980a8be47bdc93924e91436238","subject":"Fujitsu Driver: Add support for data protection","status":"NEW","created":"2024-03-29 02:44:31.000000000","updated":"2025-04-02 13:23:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be95a268a6086a6aaec1bd95a5394d1be7faf77f","_number":914230,"virtual_id_number":914230,"owner":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-26 06:53:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-29 03:00:17.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-29 03:23:54.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-29 03:35:02.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-29 03:36:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-29 04:06:02.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-29 04:30:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-29 04:55:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-29 05:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-29 05:28:19.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-29 10:08:55.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-29 14:18:04.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-29 21:09:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 14:56:14.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-05-22 00:56:05.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-14 19:22:53.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-06-15 15:54:51.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-09 01:01:25.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-06 09:03:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-07 20:25:47.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-22 04:54:36.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 10:07:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"ab82bf2aa6fbfdcef591ae7a830393a161008a64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-03-29 02:44:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a05b557524ecb469c4f8eef462919a89e8abc53d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-29 03:00:17.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1f6394b321974ad6a647754c2ed38a70\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1d3e4c7544b940dd8a9ba6a14768a148 : NODE_FAILURE Node request 200-0000002897 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"53bee5af9a9925c25999a140754283ab65425caa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-29 03:23:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 33m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8ebd4de1752e053d20cbe74cda02a84771e7311a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-29 03:35:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914230 : FAILURE in 44m 43s","accounts_in_message":[],"_revision_number":1},{"id":"73d13827666506ca2ed0e506807ae3c583b9e57b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 03:36:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2267b26d792843cf9013d4e4127e0047\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/655030b5d1994b1eb1b8e87c35d46ccf : FAILURE in 48m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/69e61a4246d949f1af594e68b81bf579 : FAILURE in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6c5ec043fa2cf424e04715dec3d3d4f93d091df","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-29 04:06:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/139/.","accounts_in_message":[],"_revision_number":1},{"id":"ffd4370cae737956bc082d28ea0a0c1b38b2c4bd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-29 04:17:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 48m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2d4c179f1307ee8f41fd35eb266a719702f5cc42","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-29 04:30:27.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/30/914230/1/check/DellEMC_Unity_FC/f98e71b/DellEMC_Unity_FC/1595 : SUCCESS in 1h 41m 41s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/30/914230/1/check/DellEMC_Unity_iSCSI/3823f6d/DellEMC_Unity_iSCSI/1627 : SUCCESS in 1h 38m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ed25579be180b40791d33dc889d6d3e4c3ed0811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 04:55:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2a888259c4246a789ccb831bc6291a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50cca2c9ffd84c3a8ea7205692188a8d : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46575cec103f4683ace3467f94e29bca : TIMED_OUT in 1h 16m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3fa31a81c324bfe9993441a8ecafd53 : TIMED_OUT in 45m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/659f0bd1e8a04b56aff8f03d8d5fca32 : FAILURE in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b1202f4bff44abe9a6ea6a2c5eb889c : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/67cf07576f264d68b1dad1b726b7e888 : SUCCESS in 1h 00m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e381f466a32c458faffb73da6d7c6638 : SUCCESS in 1h 00m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9530c14b8a8b49ee814982e17b26fb62 : FAILURE in 1h 25m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91d4e8815ab44c2f926fb9a624f1c6ad : SUCCESS in 1h 01m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05365c1f2461401ea96bd6f19cb6d13b : SUCCESS in 15m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4e257cf1d73b4e13a350b02479a4da0b : TIMED_OUT in 43m 53s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1201af1b075f49a483e855c9a307251f : FAILURE in 27m 13s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0f33e30db1884848854f6bef433173ff : SUCCESS in 6m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ac0bbe509d74c24982d0f187250cf16 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e59080de50042e899f3818e9028a6d6 : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df1451084d054ed0814e2440c167cf0d : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ebe3b9b42116479ba5eb7709da787812 : SUCCESS in 35m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/842a4e9423354dfb93a57934dcc929b1 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2ae92a3914984531ad99b0782a4d70d0 : SUCCESS in 2h 02m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6f04a93a5dcd418c812a0dae4db5f994 : FAILURE in 2h 04m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3dc1d1a5a30c4be987c9e19674d1c204 : SUCCESS in 1h 44m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/534f59e64bc749f1a8a478f1a0be7928 : FAILURE in 1h 50m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/82465be76de147ddbd30b701ada4cf71 : SUCCESS in 40m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b570df14999949b092237fc643ac4436 : SUCCESS in 1h 17m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/acacf87b652a4677a056a909c388b0ea : FAILURE in 1h 41m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bbd393f8006348cfba7d4409e7f1b789 : SUCCESS in 40m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a938a35e776c43249b07c6531fe7749d : SUCCESS in 1h 11m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/109d88c83959495793d3a3278d8ce1db : SUCCESS in 1h 45m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1331b5f4865841f59e36f088986f5a36 : SUCCESS in 1h 24m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0a88dd245db46f78811210389a2b918 : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":1},{"id":"afa07632fd27c1f83b240b14867b3d96ba89047f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-29 05:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/1/fc : SUCCESS in 0h 28m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/1/iscsi : SUCCESS in 0h 44m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e4ec1b58fda7afe832c26e9598a3bb9f2a58a8b3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-29 05:28:19.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/1/check/fujitsu-eternus-dx-iscsi/c625816 : SUCCESS in 2h 35m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/1/check/fujitsu-eternus-dx-fc/23ea933 : SUCCESS in 2h 30m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1c6382ba06447fbf6834e0102468da3866a414a3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-29 10:08:55.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-1/FC/ : FAILURE in 7h 24m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-1/iSCSI/ : SUCCESS in 4h 20m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"f702ec7433f317947d30255b1e5807c36bb5e338","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-29 14:18:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/1/20240329195639/NEC-V-ISCSI : SUCCESS in 3h 18m 37s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/1/20240329195639/NEC-V-FC : SUCCESS in 3h 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9708178c36ba85d66ce1ffaf10ee35a0a8aedfed","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-29 21:09:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/1/20240329211419/NEC-ISCSI : SUCCESS in 4h 13m 16s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/1/20240329211419/NEC-FC : SUCCESS in 8h 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"103461fcecfb806dd8204cb267b823f8f1b10fa2","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-19 14:56:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fdafa4b8b77310a49962c76fdaf9d78fcdf7e57","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-19 15:48:22.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 914229,2","accounts_in_message":[],"_revision_number":1},{"id":"4d33c2ffdc3236c27597527d01303cfd81dd246d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-21 07:06:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"da78b74db8f431a181597e461b0ef1bd8181a31a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 07:21:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61834ed397c34197968529efaaa62440\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0426d436fb3c4c95b9ae50255613a30d : FAILURE in 13m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2123cabad7b442c48b3e9789562738b9 : FAILURE in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f936ac079711d3459c431a3e3c1ad9873db2d4d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-21 08:55:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5304b750eee12f9d053e200e045cc1266c56827b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 09:19:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/142b3d6fbadb41c99a65922020de3dd3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f267a251d7024131ab58b4b14567bceb : FAILURE in 20m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b43836912e114355bdae6ace3aabcc34 : FAILURE in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f76e93dd60eeae32513552428113297607e9c1f0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 10:31:02.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4a2e01be8d18475aacb20019400c888a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b886f213d0b54f2a9c8852e839aff461 : POST_FAILURE in 31m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b3299932bcdd3937fb2e34d7ecc6103194ef12a7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-21 11:07:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 23m 02s","accounts_in_message":[],"_revision_number":3},{"id":"66a284d7d6323aabaa05b68186588ad184519c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 11:28:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/156d6d6203964b369c0608d5eefc9d77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4e9c7b3d5547439a4098ada39f8acd : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb3e7454152049fcbf01c276f6c8d18a : TIMED_OUT in 1h 06m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e10c32ea8a2345189a2b554c5d55516b : TIMED_OUT in 46m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a88434be34e4c129b1bed825fbbfdcf : FAILURE in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8cdff0581cf4aafa22cdb9155c92bbe : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/3ce3e4e4c7ab40369c40d256bf9ea4b6 : SUCCESS in 1h 02m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4671a27d9b144872a5b9ef4a4e86af87 : SUCCESS in 1h 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/616e833e10c94d78ab1d78a4478bd060 : SUCCESS in 1h 01m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68fe30bc553d4308ac58cb405620c8e6 : SUCCESS in 10m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/932e45abbbc148d593e5c86ba61fd45b : TIMED_OUT in 44m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c5054d1490584fa59adea0ff4dbd4977 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89becf7260a24e398005b4c9f35e9a9a : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a33d7c1155f4fc798f23236ecf97f76 : SUCCESS in 6m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dd0096e9c286451197b6ce108dc1a212 : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/19c4d8f3c6824628b85b6936d4c0e9b3 : FAILURE in 48m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa405971b11d4bbc881ec714543a8987 : FAILURE in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a49bf89d1364bd7be44fab9d00db060 : SUCCESS in 2h 08m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fbeba9bfb7114544a272c18ebd1be832 : FAILURE in 2h 26m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5cf3e160e1e44ade8602344b6bf68ea4 : SUCCESS in 1h 47m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/17bcb9de08e24a86a7938170b63e3ee4 : FAILURE in 2h 03m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/473b45dac2104453a6ad9f60e098d150 : SUCCESS in 43m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4f44888f6fd1440394167f74a3563cdd : SUCCESS in 1h 10m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/31be51bab6df42af85eeafd2a7de62e8 : SUCCESS in 1h 41m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/05cd1cd1e064483d990c1fcef3d9dc68 : SUCCESS in 50m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dcd024e61d174023b5235cec92f308e0 : SUCCESS in 1h 10m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8e8f5167e73d44b289fdd8b998546636 : SUCCESS in 1h 25m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75428bbf454441f291c9e55bba00397c : FAILURE in 1h 44m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b4f2b65dc274323bcca0930efae0d04 : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":3},{"id":"990b268f45ba929a8e7d8cee3d97525d5fac79c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-21 15:09:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914230 : FAILURE in 34m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b9f921bbd137a4fe6ca7b4ab6213c1de1abeeebe","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-22 00:56:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/30/914230/3/check/DellEMC_PowerFlex_3.6/44fd248/DellEMC_PowerFlex_3.6/1414 : SUCCESS in 1h 43m 33s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/30/914230/3/check/DellEMC_PowerFlex_4.5/2f0b1f9/DellEMC_PowerFlex_4.5/330 : FAILURE in 2h 03m 31s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/30/914230/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/60c502a/DellEMC_PowerFlex_4.x_NVMeTCP/271 : FAILURE in 2h 55m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"1ba49ee9cf162f4bc8e027bda89b01d02fd7f240","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-22 05:45:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11b14d3b92b816d824df8fd10e51ffb8c1887bd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 06:09:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92e62d0155ef4b8cb3911003bf2011c1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8362837b72d54652a0f9af32485ea754 : FAILURE in 20m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4c4cb0618dd34dae8c3f3348c8b60da1 : FAILURE in 13m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"767b8e5d70af02ad0a0a252332dc8511de6ffa25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 07:58:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cacdc380caf7448298dfd59d296cacc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9326fb7f42904740bc0d4b0a1aacc8ed : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39ee9dd3909345d2b06afaa0bdd5e0dd : FAILURE in 14m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23b0ede6815a4f40b7cac35abdc14465 : FAILURE in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85d4a3d867974efea2a0be71cc2a1ab2 : FAILURE in 16m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a5f2cf2dcf48c7a02d97e33f70ed6e : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/331b6ae2d911469496a2d9663b6f0483 : SUCCESS in 47m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f6cdf15e7ee47569dd0782e1ccd1c97 : FAILURE in 1h 21m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3408c3c0edc243789b84bfbea23c260b : SUCCESS in 45m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77f5f3e903c341f1bfde0eeabbdb015c : SUCCESS in 12m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/76bb6913a82044749a118fee3f8a34ed : FAILURE in 35m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b82126c43c4841e29c9c921cc38279d4 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/00004d536747448ba646931cea8c639e : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/20446b53c8c84551841b11e2fc8c48ef : SUCCESS in 7m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/49c8ef68e4354108991dd255541ad3cb : SUCCESS in 6m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/272357832db94b6ca4c415ec00cfb169 : FAILURE in 31m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b2f61950f064dae8af6477721765dc5 : FAILURE in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/26444bda2839413a8d18b3bac31549e0 : SUCCESS in 1h 53m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8044a52b5b3434ba20f5155d600a5ac : FAILURE in 1h 46m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/33a9e598177f4be1a8ed23ea915968b4 : SUCCESS in 1h 47m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/80ea16ecc44245f9aaa22b6c573636e3 : SUCCESS in 2h 02m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5f2385fc3b9049ca873caa0b09428974 : SUCCESS in 29m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f7e79bf9f05f4769868ede45f05a76d0 : SUCCESS in 1h 12m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/59625d12710f46909ac8390706bae504 : SUCCESS in 1h 35m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e6fa3ab3703d4a50b0148786323e7685 : SUCCESS in 34m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a0ecd7281701409598e2f62468f60c66 : SUCCESS in 1h 06m 25s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/71c36a751b434aef893143f11cf8ee1c : SUCCESS in 1h 47m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/514322f52120413ba056b571cacb347d : SUCCESS in 1h 36m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56a6e47afc2641d1b8f729a270420223 : SUCCESS in 1h 05m 29s","accounts_in_message":[],"_revision_number":3},{"id":"81469cf957c9cce40b2d7b28a1cf0d88cb54736d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-22 10:07:10.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/14e7c5c002c4472cae91d15e7661dc8f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8632121600e24561815d4007124f854c : POST_FAILURE in 26m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e4de0f6a056e170b329c82fe53229e32c89b0fec","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-22 18:02:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/461/.","accounts_in_message":[],"_revision_number":3},{"id":"b805b9404fe8843746501945a5330d59ace01b28","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-24 05:44:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/30/914230/3/check/DellEMC_Unity_FC/eb9021f/DellEMC_Unity_FC/1941 : FAILURE in 1h 58m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/30/914230/3/check/DellEMC_Unity_iSCSI/a5d3e42/DellEMC_Unity_iSCSI/1969 : FAILURE in 2h 04m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ea38e21ed5f4ca1656df0144f1f14f3e9d8f0bfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-24 10:36:59.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0954b66df46ba74aa01059639607ea6379ba4b5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 10:53:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb93a1bf3d884d66b239898c9ee07a36\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/216c2772ee3c43659a3c8cfd15fa7242 : FAILURE in 13m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/10192037b0f046e183a26b9ad3cf3dbc : FAILURE in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"147e8c44adbb51e6d9fad73f7a598692c9e1d680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:48:45.000000000","message":"Patch Set 4: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4bdb9dae2234584b37d63a9a019c6c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9c7b55e4b9499fa217cd61f833cd3f : FAILURE in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19827fca980a493496ca14388c071b4d : FAILURE in 25m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7295a26108a643d784df88469443dc1a : FAILURE in 10m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7b4a5d2d24748f0b9873821fdd93240 : TIMED_OUT in 50m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406e1acfd57d47b48e2ac159e0bdbba0 : SUCCESS in 14m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/25d8531a2eff46aeacbb2e2fa0350cf0 : SUCCESS in 1h 01m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/946c8ef796e94923bfffc908adabd667 : SUCCESS in 1h 22m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68003e090ba44e379c905f24121c7c97 : SUCCESS in 57m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d25d8b3dbc242dc9fa281526c8c874e : SUCCESS in 13m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b0439ae60ef44db29892b0915bef5206 : TIMED_OUT in 1h 24m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6fc0227f8e4a4657955678972d671a79 : SUCCESS in 6m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7876401a1b1341f89291cdf4b62a92be : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/730a392f54c94d93a0ca4aaefb435888 : SUCCESS in 6m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bf1f6e8e97c548ffb50ec53bdb1f8e62 : SUCCESS in 9m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b791ac5986e149f3a6cb0e074fe784ab : FAILURE in 38m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7742a1dff0e3423e9f94965f0051ed86 : FAILURE in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5905d3401d9443e9ac1f1e87ad8f1b37 : SUCCESS in 1h 55m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1765a62695ec4bcbabdcf72d7f96c88d : SUCCESS in 1h 53m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/206572bdd3b44416bed74380fc846a5b : SUCCESS in 2h 04m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ad92c600af9547feb518eb1e79afcaf1 : SUCCESS in 1h 51m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b8ee4a604b3c47a9b973dc033d14298d : SUCCESS in 27m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/caf9270e0c564971888e217ce1118ef5 : SUCCESS in 1h 12m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5b81ece2820b48cfbb1e84fee5722d92 : SUCCESS in 1h 25m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a9fe0cdd876e4f9d9ab09741482f33dc : SUCCESS in 46m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c3779371955845ffafdc01e63a8970c4 : SUCCESS in 1h 11m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/736fdf0ccdc44b338b6465a735b23a20 : SUCCESS in 1h 36m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/47a247dd5bf3446dbc800f4ba0473548 : SUCCESS in 1h 33m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8a153a36d544fe9b757c7fa74a5d8ce : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":4},{"id":"87dd9208cf5b9de39d5cc0731a32f87b2539ea7d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-24 14:11:48.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/488e698609154924b6a7b7d5910a1809\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ccf87246feb0462a8cfa0201a846bbbf : POST_FAILURE in 29m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3c7bd42c2289562d9c4cd6a39a6d268dccdd0146","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-24 17:12:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 23m 09s","accounts_in_message":[],"_revision_number":4},{"id":"cdc4f91f6eadc6fc03eb143be31895e8a8685772","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-24 20:17:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914230 : FAILURE in 34m 44s","accounts_in_message":[],"_revision_number":4},{"id":"9cb69f09b41db82dac977a3044d45b59240f6de7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-25 12:22:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/500/.","accounts_in_message":[],"_revision_number":4},{"id":"80484759a21b63cbe80292c8428a948acea0a1f7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-26 10:35:42.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/30/914230/4/check/DellEMC_Unity_FC/6285992/DellEMC_Unity_FC/1973 : FAILURE in 1h 21m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/30/914230/4/check/DellEMC_Unity_iSCSI/5756f88/DellEMC_Unity_iSCSI/1998 : FAILURE in 1h 44m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b35f0b3ecb6df539510f6cf994d2efc0e3d281f1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-26 13:20:12.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/30/914230/4/check/DellEMC_PowerFlex_3.6/7e1f289/DellEMC_PowerFlex_3.6/1454 : SUCCESS in 1h 45m 11s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/30/914230/4/check/DellEMC_PowerFlex_4.5/23f9cab/DellEMC_PowerFlex_4.5/368 : SUCCESS in 1h 59m 24s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/30/914230/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/fa5b2a3/DellEMC_PowerFlex_4.x_NVMeTCP/308 : FAILURE in 2h 55m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"623a5211b3448eda5d806276d9363ce71d845672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-27 10:12:26.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"73fdc5d176053ae3e781efebb12671001f39b9e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 10:29:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8fbb39b48004f77b27fd16453f5be36\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e232b3cdbdca462a802415863b0f276b : FAILURE in 12m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f831707fa074489b35cd965b25f63f8 : FAILURE in 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67a0bf3c4c995e4ed0238a63c4026acd4c91c768","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-27 12:36:42.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/35fc10d1bb304408b1907280ca4730f4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0c4582622a4d4eefb8565fda493a0bd5 : POST_FAILURE in 22m 10s","accounts_in_message":[],"_revision_number":5},{"id":"aecad7d9eb7289e8ddfdf9ac57dbd3c57a17d7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 13:20:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c0d52aedcba43c3aae86c130d85f9b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd541376f1694e06b12b87457ed1c458 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f21573f7925c4a3094b3413a3170c138 : FAILURE in 10m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64598e795b0d4fa5af4225e0510171a1 : FAILURE in 10m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b3cb727f69f4132be93a88625b7676a : TIMED_OUT in 1h 23m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8faa6a5d42104696853a333ee995ae95 : SUCCESS in 12m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/447c57c18ce947b986592ca136bf566c : SUCCESS in 55m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a450f842514e42e9ad0035f821ad7faa : SUCCESS in 1h 39m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6c751585a004367b492e0083b0c091e : SUCCESS in 54m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6f28bbd5d974b64965e01e7d74a7744 : SUCCESS in 16m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/19c27b434f034ff68d1a3bc2615be7bc : FAILURE in 14m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e8684c450824d16b89aef4e4d54f009 : SUCCESS in 6m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4c344a6af4354e5ea858322822941949 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09e626747d6449498aef6e9938c2c0de : SUCCESS in 5m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2d3da1f5f9b247c3a5fcef45ef250dca : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/346dcadcd3674f1d9e85a52112edaf30 : FAILURE in 32m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a1fe5ca40dd74b6ba7fe73259806a964 : FAILURE in 8m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ca41251cfef488bbf790e581facfe87 : TIMED_OUT in 3h 04m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a73799dcc8fa43b49c8d54ed22ad02cc : SUCCESS in 2h 42m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f3fd00e1686347edacb04cab00ae70a5 : SUCCESS in 1h 47m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/688273425970423c962142f216588bb5 : SUCCESS in 1h 44m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/057287ffb3c84faebda03bebbbfe2397 : SUCCESS in 39m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/233df1225921401da2fc66bab5c28484 : SUCCESS in 1h 09m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/088d248f32624c4b907687dc36bb75ca : SUCCESS in 1h 22m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aaef1f27dc794807b6a68a2eb6dfb891 : SUCCESS in 39m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a385fcc44ca34cee8bda1791350a7c34 : SUCCESS in 1h 17m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ff9adad646854d5b97421fcb985e1a72 : POST_FAILURE in 1h 44m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f5b45ec001e4fb4901772cedb167638 : SUCCESS in 1h 48m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d971fe5545d4e16a0dfb6be89c8aa85 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":5},{"id":"6eb47c83054cb7dcc1568e9f9bbd28398eb28d9f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-27 20:22:31.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 35m 59s","accounts_in_message":[],"_revision_number":5},{"id":"ed20a1189da4f9d978eff7c10cd9db48399a588b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-05-28 03:48:44.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"82f60f294b590ff654d250b99cd02491116be5f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 04:03:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed2fc4c8ae1846bbb866197a89f0f05b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fd845e6922c481abc96a2f36940a72b : FAILURE in 13m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4d1dde07877043018c308ca001db2d36 : FAILURE in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"644498e87d75b7de497eb3c77b8b06523f223d44","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-28 04:45:33.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/98d3a4d4a9a34143a77b3cb00bb33ea3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7317f12c7200420095c4eabdedf82f86 : POST_FAILURE in 23m 32s","accounts_in_message":[],"_revision_number":6},{"id":"f0fdb67e95f6aebc712fe0e6f09d5e3403ef23a6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-28 05:04:44.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 34m 28s","accounts_in_message":[],"_revision_number":6},{"id":"28b871966f83d68f5d83a9a1443f060f7352a8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 05:49:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d8332091b2e4e7c93c0a073cefa62d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e125e9b2be4ccc887211af0475b50d : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/561c9c21b27b4932b7899d6174d62b6e : FAILURE in 11m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/821c9b19a28f42a7861992a3c3e12f77 : FAILURE in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea910e4bcabc406aaef87e28b5397c3a : FAILURE in 25m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/478c73ebcd9a4d0b99820326f6b0e04b : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/77753a8b363e407fbec68f3b6a1f1ce6 : SUCCESS in 1h 02m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45d554712a6641469183b1113f1910b1 : SUCCESS in 1h 38m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9bc87b78cb648879530e8dd6fbdfee9 : SUCCESS in 1h 05m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a4afcc7103a45a0adb90cc3fb326c68 : SUCCESS in 18m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bde9923014cf465c8aac4d27c042d9d8 : FAILURE in 24m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/031fa78852a74eaba65a5a02627abed7 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cfcec7271524dc2a360a2a03d430078 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1aadb5c51cf04eeb9742e8ebab530b99 : SUCCESS in 6m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2c802adc40bb4c19904655ba7fee6639 : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a3792481be5c4db395e595d0cfcdf1ea : FAILURE in 32m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f6c0f69eaba34010bd03c2742f9b4a48 : FAILURE in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/74428afe0ba3440e866152e384b931f0 : SUCCESS in 1h 42m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0771a2691a064668999476ff2d78ccc1 : FAILURE in 1h 56m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/efc7ce6f0fa5496fbec5e7fca5118e16 : SUCCESS in 1h 53m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/feafc21ba5ad40398b0ef7f58f3cb7a2 : SUCCESS in 1h 33m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/462e79c476d94f3eb5a04e4fd34c78bf : SUCCESS in 24m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0529b9e0452c45c0aa85db3445c87c4c : SUCCESS in 1h 01m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d6ae969128b347a29f419c3f9bfea7f9 : SUCCESS in 1h 15m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e1e4776fba04884b1e0bc74f546578b : SUCCESS in 47m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8c746381f35f4b0d9ddd048b479bb45b : SUCCESS in 1h 15m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ab66aa7eb3cb4a258b55e7be8d31b32e : SUCCESS in 1h 45m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/126efb36e9144e65aecc3a4b20ba7775 : SUCCESS in 1h 39m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ca5c0c82be44e4a8fff76a27e3ec486 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":6},{"id":"f7caf358e248c9851462a9751aded5ea48522c9e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-28 16:24:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/528/.","accounts_in_message":[],"_revision_number":6},{"id":"8b2f71d496bf760728134d42ce0793097016adc2","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-29 00:00:07.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/6/fc : SUCCESS in 0h 38m 32s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/6/iscsi : SUCCESS in 0h 50m 55s","accounts_in_message":[],"_revision_number":6},{"id":"7b8d0bf0e586c08abe367024af5685a6c8a3d70c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-29 12:17:08.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/30/914230/6/check/DellEMC_Unity_FC/dc8c85e/DellEMC_Unity_FC/1999 : FAILURE in 1h 26m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/30/914230/6/check/DellEMC_Unity_iSCSI/98e3da0/DellEMC_Unity_iSCSI/2024 : FAILURE in 1h 40m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"1133c658e60b86f962133a509ba23d955b902412","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-29 22:43:57.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/30/914230/6/check/DellEMC_PowerFlex_3.6/1e2fcc0/DellEMC_PowerFlex_3.6/1480 : SUCCESS in 2h 02m 07s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/30/914230/6/check/DellEMC_PowerFlex_4.5/74aa8a3/DellEMC_PowerFlex_4.5/394 : SUCCESS in 2h 09m 54s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/30/914230/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/8876a91/DellEMC_PowerFlex_4.x_NVMeTCP/334 : FAILURE in 2h 56m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"2dbc65c8619c94fa16f0167b8965bdf0efc5ab5f","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-05 07:21:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0e1e59dacba3b583fa89ea46679769cb9f61c7f0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-06-10 23:26:35.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/6/20240611003633/NEC-V-ISCSI : SUCCESS in 3h 30m 40s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/6/20240611003633/NEC-V-FC : FAILURE in 7h 49m 59s","accounts_in_message":[],"_revision_number":6},{"id":"012a592b3238b25f9cdd4d21dbe0aec710d693a3","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-11 02:53:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"83900536fa832eb10f561fa8094992556c672b13","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-06-11 03:05:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/6/20240611033706/NEC-ISCSI : SUCCESS in 4h 55m 37s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/6/20240611033706/NEC-FC : SUCCESS in 8h 27m 51s","accounts_in_message":[],"_revision_number":6},{"id":"27bc623b8b4c696e00e8a290692b48503c450c7a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-11 05:52:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/6/check/fujitsu-eternus-dx-iscsi/05f56a7 : FAILURE in 2h 50m 35s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/6/check/fujitsu-eternus-dx-fc/3eb18d5 : FAILURE in 3m 54s","accounts_in_message":[],"_revision_number":6},{"id":"fa12be945ac5ee2e6f9b4509d2c1c5644a073c62","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-11 06:34:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7325fd0a4bd08957fc6f0841084d79182611ceca","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-11 08:37:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/6/check/fujitsu-eternus-dx-iscsi/2a931aa : FAILURE in 1h 54m 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/6/check/fujitsu-eternus-dx-fc/0f1c33f : FAILURE in 1h 34m 37s","accounts_in_message":[],"_revision_number":6},{"id":"fb96405882e0223b052e416de355852d3968f456","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-11 09:28:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d533221409ccbb8b48ae6b5338a4bd1484791142","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-11 09:44:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/6/check/fujitsu-eternus-dx-iscsi/5c383a1 : FAILURE in 6m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/6/check/fujitsu-eternus-dx-fc/3f92a3a : FAILURE in 7m 03s","accounts_in_message":[],"_revision_number":6},{"id":"bf5daaecd4b915529c76a15fd469d563c0098e00","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-12 09:57:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a63b34fc7927bd3aa60668a18b06dd545a98bf0e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-12 12:21:39.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/6/check/fujitsu-eternus-dx-iscsi/55fef2a : SUCCESS in 2h 15m 48s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/6/check/fujitsu-eternus-dx-fc/e625848 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":6},{"id":"9b664a61b2a39047e17b3bd849edb0a8cc6f535a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-14 09:11:22.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"29c8cb9fe9bd58bd44a4296d91d93581192640f8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-14 09:14:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"153c9a093a3f44e0ac841d96afa0d840a4d30370","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 09:49:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb321d3a46534a089ed34d7d6c00da55\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/198af716360f46b1b723fc2f82c0ac8f : FAILURE in 35m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01be2a7ca3ac44dab634beded5c79d2f : FAILURE in 16m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b0a0d26b6101129a7e43675bfac4ae5d268798c5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-14 11:26:29.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4679fec00b6d44c8a4b47e5263dd69b9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/24299a5f3c7b4f30a9c5e6f5f081d9eb : POST_FAILURE in 30m 11s","accounts_in_message":[],"_revision_number":7},{"id":"f3de1b4ad3dfc38342f1b698a431f4e45b36de07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:00:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80453375f8a34df58263d54bc17fca1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0631e5cec5db45708eca79ad78a3a719 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ee9c18873764fc38120b20d481a1b29 : TIMED_OUT in 59m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b75612856bbb437389eead091a5b74cb : FAILURE in 13m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8b7aa44bf7842c38e5466f4dea75070 : FAILURE in 26m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7335f46cd24642a1b803b4010cfd3531 : SUCCESS in 13m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/78b139963c0f4325ac67556769c9d947 : SUCCESS in 1h 02m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c82f2d3450d8445991463f440fcdc822 : SUCCESS in 1h 26m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99ec0e01ccc44a3a99d5ee80a7967e23 : SUCCESS in 1h 04m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3094ced66d114e55aa976e6b45c254d0 : SUCCESS in 20m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eb30a8d80e104215999a01b08b3b95c9 : FAILURE in 30m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/51363018a4354478aee5a8810540fd91 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/44dfd4ff735d4f02891c747fcbd0f811 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6ce4f805945435cab61402c94b4c6b0 : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/029ff52b145749aa9870bebb5b4d0f4a : SUCCESS in 7m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7019c711478b4c4a85254e13399098f0 : FAILURE in 42m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d92ef247dbfd47e6bf78a5ae657d9ca9 : FAILURE in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/45c90d803dd0423b9b9986cdcd574df7 : SUCCESS in 2h 11m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f21b521c8e64946ab850b6277a3aa58 : FAILURE in 2h 14m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b7dc5947b244a80aff545b5ea69e314 : SUCCESS in 1h 58m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/43303444650c48089d15bd19d178ccae : SUCCESS in 2h 42m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b31b41545e1b459fb4de12c3a4fad414 : SUCCESS in 29m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/69d7e700f9674372b9fcbe8964324c5a : SUCCESS in 1h 14m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b750327afeb24d6288cbd6fb31ccc729 : SUCCESS in 1h 28m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3c3a45e4477b4fb1920f16336158c020 : SUCCESS in 43m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2530bebc90b74e90b64320d0b97c3570 : SUCCESS in 1h 35m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ecb06973b57b4f508e49a843d15dddf7 : SUCCESS in 1h 37m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70fe759500ad4554afbaa64dd4631d13 : SUCCESS in 1h 37m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75c961a3db714ef1bc02e4ecafacae36 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":7},{"id":"132425714928aafecff1cbe44d5bb1e314a07154","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-14 19:22:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/914230 : FAILURE in 34m 31s","accounts_in_message":[],"_revision_number":7},{"id":"7c805a40d66b956eafa2a98ba79a689dd738752a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-14 21:24:24.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 34m 28s","accounts_in_message":[],"_revision_number":7},{"id":"6562ee777878002c3e31789369cc479db2155e50","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-15 08:58:58.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-7/FC/ : SUCCESS in 12h 57m 44s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-7/iSCSI/ : FAILURE in 13h 6m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"9d49fc2a1c8b4b106929bf7cb53af144ad8f8c20","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-15 11:27:38.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/7/check/fujitsu-eternus-dx-iscsi/ee19fa7 : FAILURE in 2h 34m 15s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/7/check/fujitsu-eternus-dx-fc/42543fc : SUCCESS in 2h 36m 01s","accounts_in_message":[],"_revision_number":7},{"id":"a9634969e12bf998a7253d8cda0491b763346974","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-15 15:54:51.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a4b32840d310459fbd6821668050c54b : SUCCESS in 1h 18m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/659dc9320dac415781352ea15f10801a : SUCCESS in 1h 16m 24s","accounts_in_message":[],"_revision_number":7},{"id":"79055dc09d866b98554e7958bbc12ebf03d94d47","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 16:53:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/125/.","accounts_in_message":[],"_revision_number":7},{"id":"18e9b871c282ccd69cff1a5121f1859403c4df95","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-15 17:52:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/611/.","accounts_in_message":[],"_revision_number":7},{"id":"7efc22df7e7a0166b894cfbcc75c49ba206ed7dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-16 06:35:34.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/30/914230/7/check/DellEMC_PowerFlex_3.6/dce45b8/DellEMC_PowerFlex_3.6/1586 : FAILURE in 2h 07m 03s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/30/914230/7/check/DellEMC_PowerFlex_4.5/d01af79/DellEMC_PowerFlex_4.5/481 : SUCCESS in 1h 47m 29s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/30/914230/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/b1fb904/DellEMC_PowerFlex_4.x_NVMeTCP/431 : FAILURE in 11m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"b9b8666c55875081e3ad14af04b7fc076a1ab0fc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-19 07:26:28.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/30/914230/7/check/DellEMC_Unity_FC/38bacb2/DellEMC_Unity_FC/2095 : SUCCESS in 1h 23m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/30/914230/7/check/DellEMC_Unity_iSCSI/5708783/DellEMC_Unity_iSCSI/2130 : SUCCESS in 2h 11m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"917c38c1cfb96f1f522be54f0c3c2752b5216373","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-06-21 02:31:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e6544aa14f1a1fee824305b42aa102ecdcf27cd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 03:07:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d034b6e46845a6841f86c6a0b2e8cb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3dad7c0b96ae4b138eeba69121a27c3c : FAILURE in 25m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44d4fd403bb54947bf0ada7423fe1fdc : FAILURE in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8b3b40cb0e106b63be85f9e3824e85eb4b39c366","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-21 04:45:07.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/cb09130f3d5a4f06ba1586103d2c4580\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/86e3c6382cc8442b982fde67d8d04a31 : POST_FAILURE in 24m 25s","accounts_in_message":[],"_revision_number":7},{"id":"0b9f081e8c99302a379bd5dc5d5bf4ffa207d0e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 05:57:09.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1781acc2961469bb1e03bb699f5eb85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093379990a7e48668004f73da1aded75 : POST_FAILURE in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44a4f68df2e044559fbd290db2e33f3f : SUCCESS in 11m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c090cc3262154634ae7bddc483d1d20a : TIMED_OUT in 42m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1262fdc761e24b8693ae7100287613dd : FAILURE in 27m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15560e9d2b714020b6f737ee6f492e0c : SUCCESS in 16m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/c20343a008d242a294a8fdf623c45213 : SUCCESS in 59m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd6440a2b68c4a5cbc87abffecbf81d7 : SUCCESS in 1h 37m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82115e701039419eb48100a8f3544a42 : SUCCESS in 1h 03m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/844b2094598749ce9b2f98ae9b214245 : SUCCESS in 18m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1cef4ede2f5f4b998512bcebd647463b : FAILURE in 26m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/069db86384224010a5a25f8124d329d7 : SUCCESS in 6m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e17420a342604e76b30d6f34e8cca72c : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0cd76c19078e49f6b92847f89fb54d4a : SUCCESS in 6m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d792cdb3ed76476eabbf476c16284477 : SUCCESS in 9m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ef3fdf86f22d4bb29c46f002a4229fb2 : FAILURE in 46m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2cd367ee4f1c4e558ade9bc3aae34475 : FAILURE in 7m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3008ef4554204d7db23a21019aaa8699 : SUCCESS in 2h 26m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/99757f36e534438397971eec84c56487 : SUCCESS in 2h 27m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/53437c962e3240b4a99223a31860e445 : SUCCESS in 1h 55m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4dacefd6429547f7a29b0275aa32f99f : SUCCESS in 2h 26m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c68c98b9299e4d58af0146b48bd2362f : SUCCESS in 43m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dacc41b25a1f45bc9f16c49f80adad15 : SUCCESS in 1h 10m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1155985e6d35407bb107f8dff5438e38 : SUCCESS in 2h 04m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/353c1f690cc1484f818a52a497f45003 : SUCCESS in 44m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b9986d71805948af98eb21116d66a9b2 : SUCCESS in 1h 31m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/aeccd78de05545eea95649dc9c2f8dc8 : SUCCESS in 1h 25m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f0b09df248664318a9cb42a9c7783426 : SUCCESS in 1h 52m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b5be02cce5a44c8b925e9d3963480e2 : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":7},{"id":"f5d5e368ffedac36272f2260d0bf1b7862eb9de4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-21 06:48:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9ea05bf8dc444dbf9e73ca4a1d79aed9 : SUCCESS in 1h 20m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/039350a5416d45da9c04bbe2ecb06567 : RETRY_LIMIT in 8m 00s","accounts_in_message":[],"_revision_number":7},{"id":"368421065a91e69e6bb11153a565bb93ed684a24","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-24 15:16:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c38ed624109943b48ab52fbfcc4a5e6f : SUCCESS in 1h 20m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e946ccb2f5c545bf8510622a564798d8 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":7},{"id":"e108847c99a5e5953b13f09189e8b66113a3c504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-12 08:51:42.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"20a890c85cbdb3ef99a28e02e6ce3ea6af29e3b3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-12 09:08:01.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":8},{"id":"7951a7028d93bbc04d95c1a7e0708550aba29dfb","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-12 09:08:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/914230 : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":8},{"id":"fbbde0b6b55d28247a248e3fd959dfac8ae074db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:29:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4fb9a112ce747f2abbece5532b93b75\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21d6217eaa7f4554b784426ebbf35072 : FAILURE in 33m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3dc136ef73e4fd89ebd55e333c1ddf2 : FAILURE in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7a855e6cc4bb40ae9489fbe18107d5ecd9c32c3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-12 09:53:59.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.","accounts_in_message":[],"_revision_number":9},{"id":"256bc262de310db46b86ac57899e0d0b55bd6ead","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 10:40:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f180e2de7ef42c9ab55d888a0a74b07\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfddc0739f644cc4a395a10952b5797b : FAILURE in 3m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/919385b82c6744909a5a7856e5294f5e : FAILURE in 30m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d89f3ce27f7a5ccbdcd74b108146c15a20c136be","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-12 11:43:05.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2a22a813ad204bea9d8ce3f381295fdd\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ec8874dcc02348a89274dd6ed2978f0a : POST_FAILURE in 23m 58s","accounts_in_message":[],"_revision_number":9},{"id":"50f30c60cc5a825c681cbbcf8d132798e1ac423b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:21:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a78445d12f94ba6ac84efd560679a56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c364f9e60d984522a941a2d3f12a4b89 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a951625f9cb7451cacb35f04036d39b0 : SUCCESS in 11m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/909fceb0b3694221bea84c85576d510f : TIMED_OUT in 1h 16m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a987fabcd0f4066b269b936a8086e20 : FAILURE in 24m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e30e7b5fea943caa6c5abfef50f1121 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/a1df534a63614ad6ba7122c791d0db22 : SUCCESS in 57m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6909a45109874172a05e7fbbe1062dbf : SUCCESS in 1h 22m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df551d8dd91f4fe39236a48c7f3ca8dd : SUCCESS in 50m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/225ba498900448a2a133e9db1b5afdfe : SUCCESS in 16m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4b97a50c95cb41dbaeb0b45da1f5f747 : FAILURE in 28m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8573a72dd65f477587c014b42a961dc1 : SUCCESS in 6m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5b4d09b4c7d48d99d2f62707d5f29b5 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee060326f1bc4b668a1ed5f547506e39 : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bfaf833dd3b1491987eecf2bb60bb1c2 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ee5a4fef61d641bea79158c47d4a23dd : FAILURE in 37m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/353d4d98a9324b05aaccde93c64f9217 : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e3cb004e186a4e5ca20e0ae87bdc8113 : SUCCESS in 1h 47m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/65e6098aba97464985fb46a9581d12ee : SUCCESS in 2h 19m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/27152f7dd8a149548c83ce32539f2a87 : SUCCESS in 1h 42m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ebf6931f54f74e9fb6641b48f541e61a : SUCCESS in 2h 00m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/56f7659f2b154b02ae6154152c0c4d2a : SUCCESS in 26m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de23c105bf50475587816613d0eb6cfb : SUCCESS in 1h 09m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e0cc00ae96cf46e195d573529a320d7f : SUCCESS in 1h 29m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e2e4531814414234ba201897012a84e6 : SUCCESS in 45m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cfe6c103de6d402387142e9bbaa0b7ad : SUCCESS in 1h 15m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6245e754461b431092c8dc5f338f15f8 : SUCCESS in 1h 52m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40fd8b476df04250a9a7cdacf9be8fd0 : SUCCESS in 1h 39m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/626ef340a00a4d089ed919761b4f5fc3 : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":9},{"id":"d3158583e424522c39f9e3af53e8192e36098dd9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-12 17:00:07.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/9/fc : SUCCESS in 0h 39m 20s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/9/iscsi : SUCCESS in 0h 57m 53s","accounts_in_message":[],"_revision_number":9},{"id":"0e4c8bd5b7c1209f24ab90f6e60e7c1b1730c981","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-12 19:00:35.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5944c6c52097457ab51e326f89061af1 : SUCCESS in 1h 19m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/66d6df8acb8c4fc8a08972106a8d9b18 : FAILURE in 1h 13m 37s","accounts_in_message":[],"_revision_number":9},{"id":"709e9ce5203d3d88d8209ebc4b56bca848d9ebeb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-12 20:05:02.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/9/check/fujitsu-eternus-dx-iscsi/8016b24 : SUCCESS in 2h 00m 59s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/9/check/fujitsu-eternus-dx-fc/a7485a6 : FAILURE in 4m 04s","accounts_in_message":[],"_revision_number":9},{"id":"4af39d757943ee182cef6fe0f924269d38408be3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-12 20:39:50.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-9/FC/ : FAILURE in 7h 52m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-9/iSCSI/ : SUCCESS in 4h 7m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"1c3d2d0e227af2234ac19f653ac6e43564fb0a34","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-12 22:57:18.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/30/914230/9/check/DellEMC_Unity_FC/7458f5c/DellEMC_Unity_FC/2219 : SUCCESS in 1h 23m 30s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/30/914230/9/check/DellEMC_Unity_iSCSI/b645d44/DellEMC_Unity_iSCSI/2228 : SUCCESS in 1h 48m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"ae2c49b1a6b25fb5bde0111ef487104de91f36b1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-12 23:39:34.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/30/914230/9/check/DellEMC_PowerFlex_3.6/92d9478/DellEMC_PowerFlex_3.6/1704 : SUCCESS in 1h 50m 14s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/30/914230/9/check/DellEMC_PowerFlex_4.5/7adafcf/DellEMC_PowerFlex_4.5/585 : SUCCESS in 1h 56m 15s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/30/914230/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/7a4cd87/DellEMC_PowerFlex_4.x_NVMeTCP/539 : FAILURE in 13m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"471db2fb9baa355f2b672e6105541d89c3d892ad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-13 00:36:13.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/228/.","accounts_in_message":[],"_revision_number":9},{"id":"2a9b567a1da34750c08ae484d7d9a5cf9df941a7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-13 15:41:35.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/9/20240713161028/NEC-V-ISCSI : SUCCESS in 8h 07m 24s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/9/20240713161028/NEC-V-FC : FAILURE in 8h 31m 04s","accounts_in_message":[],"_revision_number":9},{"id":"5b6487c5cab06617ee28598e54fddc1a9f09ec00","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-13 18:23:04.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/9/20240713184751/NEC-ISCSI : SUCCESS in 5h 21m 08s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/9/20240713184751/NEC-FC : SUCCESS in 8h 35m 09s","accounts_in_message":[],"_revision_number":9},{"id":"ac204822a87b22468355c78793056fcc89bfdb1a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-15 11:02:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/725/.","accounts_in_message":[],"_revision_number":9},{"id":"08b3749c46c442c73d65d6b39fea878cba8187c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-08-09 01:01:14.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"46dfd6c1eca5fcab3d47ef8b1dfe1a2aea52becb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-09 01:01:25.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/260/","accounts_in_message":[],"_revision_number":10},{"id":"c2ef9e189b8e681192fc6787dbcc7f5742a3647e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-09 01:01:28.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/260/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"e42a88f5dc48e1fff6c36af29e2826f2ac291ba2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 01:31:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e33fa97c5e744c7a854d7652cd29bf09\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eed315549c654c95bbacfc23d755d4ad : FAILURE in 26m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e395fee3bf3e46c9af34a55d5bf9f5f0 : FAILURE in 27m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8156189b59820a00650d51b06908c9f4d5836e7b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-09 01:47:15.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/914230 : FAILURE in 32m 15s","accounts_in_message":[],"_revision_number":10},{"id":"5d83607e3567670880b68acbe0bb349be91f3083","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-09 01:53:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 34m 20s","accounts_in_message":[],"_revision_number":10},{"id":"7a52e79104294e9cdb54f4726ce81492f7045533","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-09 02:22:55.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914230 : FAILURE in 34m 51s","accounts_in_message":[],"_revision_number":10},{"id":"dd404fb439555f9098cd37e7ee12ef782040277d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 03:34:45.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/692a059594af429a8b4ec2fbdd96ae77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd5778a59fb4d3da86bb3dabf6b7304 : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b5d592d41f24c329278d05f9ca2ebfd : TIMED_OUT in 41m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/332630aecd8c4d46962f432db9660c42 : FAILURE in 11m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4564d72097246419ddd5529685737f9 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a568f73562415193d3bfcf92a4c77f : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6111694d99b44910a6ef0e72ca5a8fcd : SUCCESS in 53m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72f35a2e6e54476c9eeaf10b78be1759 : SUCCESS in 1h 16m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc58d0f05975467195bf6a1b1f599075 : SUCCESS in 46m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d776ac016684e9e9e77dc28e25bfe3a : SUCCESS in 15m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/27deed5148d74afc946ae636365b0bda : FAILURE in 25m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/93b1e7fa05804669ba4d552315bd1263 : SUCCESS in 6m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da52333db80f48dfb6cd3794df0beb88 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8cd4d2f471114a0fb7b02f83cae52bc4 : SUCCESS in 6m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cae51acb5aec4accb0844b35eac9313f : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8b85896609ee42838f766ff2f27e5887 : FAILURE in 35m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ed348c27854470b998f2f1b97b737f9 : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ffd5be9cb75c45c8ab9619ef0b14759e : SUCCESS in 1h 46m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/137e1c4c7e744d629fba869538cec933 : SUCCESS in 2h 25m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dcf4bed0695e4d9681b67d111f6bc26e : SUCCESS in 1h 51m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0ed661b281c14a8ea3161009b9e3d474 : SUCCESS in 1h 39m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/511f46b16040415b9206fbf997d63c3a : SUCCESS in 37m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ec53a88b4b014492a2e9b3565e515691 : SUCCESS in 1h 20m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ff3ed3089784be5bfba4b7598315635 : SUCCESS in 1h 16m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e457b5d9a4f0429a82b1ef526f8255a4 : SUCCESS in 46m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f91044b2dadc449d8c82db17ecc27dc4 : SUCCESS in 1h 26m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7d68058bf97345308e3ba161099308b1 : SUCCESS in 1h 18m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4209ab6ca99b48a0bac2006594ddc935 : SUCCESS in 1h 54m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61c86211fb3b44b79c124c08c8b56ab9 : SUCCESS in 49m 26s","accounts_in_message":[],"_revision_number":10},{"id":"47d5f16bf10f0ca9091ed1ed92f5babb24158a61","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-09 03:47:18.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a884105f88824515b7b4e56f7b79b73a : SUCCESS in 1h 14m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/062266b4655843e9b31ab80ad7f9d90b : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":10},{"id":"5c909d0e934703a2fe88bfc60f191cdddc929bf3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-09 04:22:49.000000000","message":"Patch Set 10:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/10/check/fujitsu-eternus-dx-iscsi/560874c : SUCCESS in 2h 47m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/10/check/fujitsu-eternus-dx-fc/4d41166 : SUCCESS in 3h 06m 14s","accounts_in_message":[],"_revision_number":10},{"id":"1e47d7865d82360529c246aa007741fab2cd7287","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-09 10:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/10/fc : SUCCESS in 0h 47m 45s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/10/iscsi : SUCCESS in 0h 53m 32s","accounts_in_message":[],"_revision_number":10},{"id":"b04817e916d6161b26432a13c58b640892a645f1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-09 11:59:22.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/30/914230/10/check/DellEMC_Unity_FC/2f72767/DellEMC_Unity_FC/2396 : SUCCESS in 1h 51m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/30/914230/10/check/DellEMC_Unity_iSCSI/d4a518a/DellEMC_Unity_iSCSI/2401 : SUCCESS in 1h 54m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"f715f5cfdba8e08dbff4180b3a680cec9795235a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-09 17:01:20.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/30/914230/10/check/DellEMC_PowerFlex_3.6/af6bbf3/DellEMC_PowerFlex_3.6/1882 : SUCCESS in 1h 54m 36s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/30/914230/10/check/DellEMC_PowerFlex_4.5/61c557a/DellEMC_PowerFlex_4.5/761 : SUCCESS in 2h 07m 01s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/30/914230/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/c5286ff/DellEMC_PowerFlex_4.x_NVMeTCP/730 : FAILURE in 13m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"5b58582f9c6544cdb828b4d0d738718417de4b93","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-12 00:14:55.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/10/20240812011954/NEC-FC : SUCCESS in 5h 46m 56s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/30/914230/10/20240812011954/NEC-ISCSI : FAILURE in 7h 54m 56s","accounts_in_message":[],"_revision_number":10},{"id":"eb04a8b85e6e92499634c637733f611a89869fa6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-14 06:41:43.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/879/.","accounts_in_message":[],"_revision_number":10},{"id":"500f6b5fb2d1a61166b9daa7dd81360edfb273ce","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-19 02:50:43.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/10/20240819001927/NEC-V-FC : SUCCESS in 4h 09m 27s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/10/20240819001927/NEC-V-ISCSI : FAILURE in 11h 31m 13s","accounts_in_message":[],"_revision_number":10},{"id":"0cc0316c2306519667dbac03b287b123d355cfb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-09-06 07:21:45.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"22b7b7cf58df605355bd6d525c045a8808f157b5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-06 07:21:53.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/490/","accounts_in_message":[],"_revision_number":11},{"id":"70d0693b0ad70c29c3d5066b0db628f4c3685575","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-06 07:21:59.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914230/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914230/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"6646336d81318826a00a69482da3a43b7e8e93d3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-06 08:01:26.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 35m 22s","accounts_in_message":[],"_revision_number":11},{"id":"dc6e028fa364bada038a73f4bd5b2b3df20557e3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-06 08:03:34.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914230 : FAILURE in 36m 45s","accounts_in_message":[],"_revision_number":11},{"id":"57af4256aabf77c1228e568c9e4caf20da7fb76d","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-06 08:07:58.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/914230 : FAILURE in 32m 32s","accounts_in_message":[],"_revision_number":11},{"id":"76a749d7f221e31d2cf7b1e8274550015970a6db","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-06 08:14:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/493/.","accounts_in_message":[],"_revision_number":11},{"id":"cec99aaf12015721a0f8fd7465670a56d2237080","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-06 08:42:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1081/.","accounts_in_message":[],"_revision_number":11},{"id":"7c518dd66bf544b83b50f1be585a8167cb8d7581","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 08:44:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56c79ffc04da443db219a3431fbf7c56\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c9a29309320438895189ba205c17de9 : FAILURE in 30m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1af1cac8b874b0aaa3d86f5e1fac07e : FAILURE in 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9a2833625ca411551598cb859672e1edc51d0f64","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-06 08:44:54.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5ef87b4b7bda4297a560f78067c5124f : SUCCESS in 1h 09m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3df2b59bc0ca46bf915bbb4de9b44258 : SUCCESS in 1h 19m 33s","accounts_in_message":[],"_revision_number":11},{"id":"b5dae5e9978ac98929b067a52d3a238289675549","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-06 09:03:56.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2477029869424bb0a1e99c4753d10101\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/97343b62b1004be8b2d4fbe9ca23cfff : SUCCESS in 38m 28s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d5969fe039514f9cb7204d07896aed26 : FAILURE in 1h 24m 53s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6cf37a044d21420d9817121faf3650db : FAILURE in 1h 40m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e13c250fd82a462bbd2d8b108303ec65 : SUCCESS in 42m 58s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"036c408610cb2d1daf6e533b87f271065b83e220","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-09-06 09:13:47.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/11/check/fujitsu-eternus-dx-iscsi/8d1ee0f : FAILURE in 1h 40m 26s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/11/check/fujitsu-eternus-dx-fc/ad10621 : FAILURE in 1h 46m 05s","accounts_in_message":[],"_revision_number":11},{"id":"4aaa0d873cf01c520cef16c2676dfb1e608bda46","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-06 10:24:55.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/30/914230/11/check/DellEMC_Unity_FC/37a49dd/DellEMC_Unity_FC/2564 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/30/914230/11/check/DellEMC_Unity_iSCSI/6bd77f5/DellEMC_Unity_iSCSI/2569 : SUCCESS in 2h 10m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"a8a1dfd48e9fd4d36a27e3d2f7219dd3974c711f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 10:33:28.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b70b828f353d4d0da844e2d5bbeff067\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72e3614a2704773b7566a4f07d9c8e6 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfc15b0e04434baa9891ac746ae34549 : TIMED_OUT in 1h 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3ed8226489b4e7e914377befe42c308 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02175cf18d734c7498b78fe5013bceb5 : FAILURE in 12m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902cd4a47ad64987bb36ec414390c14e : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/c2672fe1e0464a1fa71763f018a1c19a : SUCCESS in 44m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/974c78cede8d4b3d90e65580bc3bc692 : SUCCESS in 1h 15m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ab8f12c01434341b45ba7b30fa56367 : SUCCESS in 57m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68155145b1414ea2afd18d75f06b02bc : SUCCESS in 18m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fbb2a451444647b9931ef45f648fd20e : TIMED_OUT in 1h 51m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/45d1264fac0e42538efd1f53f9302756 : SUCCESS in 7m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/830a716f296c4d22be99173023c73483 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a5b8dd4ad0154dd497ccfd452bfa02ce : SUCCESS in 6m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/594c413279b141dc81067f5b63da259a : SUCCESS in 9m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/06802fa62c5548be9419c0c20b3f2bb8 : FAILURE in 37m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/223dd55faebc43038013111723743300 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/88ccf457897c4472a3d99e5586a9cea1 : SUCCESS in 1h 51m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e68a0bce68be40bb90e969ebbdbb5a37 : FAILURE in 2h 47m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fe406a1586584be09de7417ad1be0a6c : SUCCESS in 1h 49m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d2b39f579b694d6db2dade71f35ff71e : SUCCESS in 2h 07m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/47d24bc721144e25819e605efc2541ce : SUCCESS in 36m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/edfcfe101c96471fa6de8df9e7e6d4e5 : SUCCESS in 1h 09m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/292d0f2eec4244da893806c040334928 : SUCCESS in 1h 18m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5c7986a160f146cbbc2f0113448f5fa8 : SUCCESS in 36m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4441ac25d42c418d896fe5dd1c70308a : SUCCESS in 1h 24m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4c25d2d3501f454bb440ef48a9e2aa41 : SUCCESS in 1h 32m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/824c39c085c447aa89418f48259249f7 : SUCCESS in 1h 18m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3202a7e7daff491fa0162db786507e60 : SUCCESS in 45m 02s","accounts_in_message":[],"_revision_number":11},{"id":"90146203e72b890245856e92169f2e46798c1d02","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-07 20:05:21.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/11/20240908012852/NEC-V-ISCSI : SUCCESS in 3h 27m 18s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/11/20240908012852/NEC-V-FC : SUCCESS in 3h 36m 26s","accounts_in_message":[],"_revision_number":11},{"id":"414942a804faf6fd2a683da665119f0dfff41eef","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-08 16:59:15.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/30/914230/11/check/DellEMC_PowerFlex_3.6/53d9773/DellEMC_PowerFlex_3.6/2059 : SUCCESS in 2h 09m 29s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/30/914230/11/check/DellEMC_PowerFlex_4.x/f734c63/DellEMC_PowerFlex_4.x/942 : SUCCESS in 1h 52m 20s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/30/914230/11/check/DellEMC_PowerFlex_4.x_NVMeTCP/0a1c9ba/DellEMC_PowerFlex_4.x_NVMeTCP/897 : FAILURE in 5h 51m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"3e0fa5cc52e83078e46d865007329ef29f1ed651","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-08 18:41:26.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/11/20240908215414/NEC-ISCSI : SUCCESS in 5h 19m 16s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/11/20240908215414/NEC-FC : SUCCESS in 5h 47m 09s","accounts_in_message":[],"_revision_number":11},{"id":"24f5e629be3caaed35950f0f51796f819b246b2e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-12-12 14:21:57.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7c3a5064aa7792540bfda25d01edc03563936ede","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-12 14:23:28.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 910481,8","accounts_in_message":[],"_revision_number":11},{"id":"ea42a7e430f6e32d59c4d8e7ee73bf518fd3b621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2025-01-07 06:35:52.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a690b6e30c5940e0eba9a3c8134188cb8800a42c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 06:35:59.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1157/","accounts_in_message":[],"_revision_number":12},{"id":"ff6e4976b5da7aca019f9d41a5bce0581e8c6404","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 06:36:52.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914230/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914230/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"9654d72a8c5b7c81c8c9e4c6abdf300d282d1f9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 07:21:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f3f189a9f3f4e9490d9301a3e2c891b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c647b6c7d464aa0b019a69fd0642373 : FAILURE in 29m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69b0e3df1c3e4198bbd1828bea5c4d34 : FAILURE in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"11db4a4611cfeec845fc6575e5525dce90cb7e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 09:20:27.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc1a36032da047ee8521a919b087f4d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb667724c0f43b38390a312d8ecfdae : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/373c53752e1d4a7281f0ac318eee4087 : FAILURE in 39m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f69d9adcaada4e399f42954b65be78f8 : FAILURE in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8585a301dc304f419fefd3e80c45335f : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/a078659d06ab4ec0b12e05bc0b7bf558 : SUCCESS in 1h 02m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/521a903cb3f646218aec55822ab0787b : SUCCESS in 57m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e40375df12a341ef90f899aec73b33c8 : SUCCESS in 1h 29m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/021e7aae9d29486cbc8a3264e7fdfc35 : SUCCESS in 1h 09m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1560599e74af4336a570bce59fbd487b : SUCCESS in 16m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f75723f9f4234846a5adcc157fc51b8f : FAILURE in 32m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/13560fa796f544b895b12bceb5f68a36 : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e727445016694ed3a597eff78218c550 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d84adee81342481ea2bbc4a1fc74863e : SUCCESS in 7m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8edbd2376c6240d686e3e593b52e7385 : SUCCESS in 9m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bc5a4df1a3a34416a92e16c4c011a8d0 : FAILURE in 27m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b68f262852c548048bc36b37737fcf65 : SUCCESS in 7m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/43e67f1d5d1b49b4845a2ccd6f5e2951 : SUCCESS in 2h 03m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f4ecbaf821a54e2daccaed918d160c61 : FAILURE in 2h 05m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/11bab7caa76d45f89992ba949bf44ce2 : SUCCESS in 1h 42m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/98a9d3dd039c43eea1fbbea996fd449d : SUCCESS in 1h 56m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c7288b1587ef4b7c988fe15d91ff3f6c : SUCCESS in 42m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6633fbcfaba044ffbe6029236190ddfa : SUCCESS in 1h 13m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/50cae8329c6d4ed9880c023aa561594c : SUCCESS in 1h 36m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9bcbf5ccb90346bbba4629dc3ad5aeb4 : SUCCESS in 54m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/811cdd26e064482297615ac1b0e3df80 : SUCCESS in 1h 52m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0d87460cf514475a83c468c69c067eef : POST_FAILURE in 2h 35m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4c5125b28394fe2bdbd98a28167aa0d : SUCCESS in 1h 47m 53s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27b85f10b8cf4c468b32fa2ece9cd1e7 : SUCCESS in 1h 29m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea0052c4f682408b9bb8fd79e5ec601e : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":12},{"id":"f5443ec53962dd9e005998210905b2d1d5351530","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-07 09:42:12.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/aa30a7f3ac4d47788426add4b98420a4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5818a3e7cde4b3eb16274c8953f3058 : SUCCESS in 42m 16s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1fdc7b426c0042178235583be2e980a1 : FAILURE in 40m 39s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e71775a818a849a9885b1b1f7b857651 : FAILURE in 1h 43m 54s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e36f5431f6044b14aa6431c2a9ccad05 : FAILURE in 40m 09s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/55cd916f480c405ab3de10fb2288d2ac : RETRY_LIMIT in 44s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dae47c7dbd84490c83bd6769f3b3ad3a : NODE_FAILURE Node request 200-0000021224 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ff91d2a3ce1433e8fe75d5ebf8a7ade : RETRY_LIMIT in 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a1dd9257318e48aabe70596e042d6305 : RETRY_LIMIT in 44s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"07a03d63972f0b2725f4b5f26e3b249ed9648975","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-07 11:03:57.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/914230 : FAILURE in 34m 29s","accounts_in_message":[],"_revision_number":12},{"id":"e1b4c2514e380036406d46d2b469862fcc61d183","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-07 11:32:33.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914230 : FAILURE in 34m 51s","accounts_in_message":[],"_revision_number":12},{"id":"6ce523d50cb4a6d9717fb4e07b9ced8332e0cdac","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-07 11:40:30.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 36m 26s","accounts_in_message":[],"_revision_number":12},{"id":"101075978370caeedc3ece4559c0ea7673efda66","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 14:30:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/195/.","accounts_in_message":[],"_revision_number":12},{"id":"bccc0681a8f85dd56fb1e9b58fe7f755e1432454","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 16:31:23.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/184/.","accounts_in_message":[],"_revision_number":12},{"id":"7486cbcf2114e024b8d8f0370578facfdc863ffa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-07 18:41:54.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/336846df0f8a4920ae23914c16434d13 : FAILURE in 1h 14m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/06a12791912d408083513143d2f66b27 : SUCCESS in 1h 17m 54s","accounts_in_message":[],"_revision_number":12},{"id":"25e0973f468bfe2c96b96cfac39f0258ed6b8b20","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-07 20:25:47.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/30/914230/12/ : FAILURE in 1h 15m 35s","accounts_in_message":[],"_revision_number":12},{"id":"ebb48f3801524a005df4ce62354147c354c18b0d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-11 20:47:26.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/12/20250111203955/NEC-V-FC : FAILURE in 8h 42m 48s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/12/20250111203955/NEC-V-ISCSI : FAILURE in 9h 07m 28s","accounts_in_message":[],"_revision_number":12},{"id":"535fb750e45ff3f05307f3d081df8654b32c2263","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-12 00:00:09.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/12/fc : SUCCESS in 0h 48m 44s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/30/914230/12/iscsi : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":12},{"id":"cedc26cafd81a606bf680bda92dcc9f39e06b4de","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-12 20:09:04.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/12/20250112203043/NEC-ISCSI : SUCCESS in 5h 19m 59s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/12/20250112203043/NEC-FC : SUCCESS in 8h 38m 19s","accounts_in_message":[],"_revision_number":12},{"id":"2e2722b64e9009dd5e5b01f2c3f80dcfbc613131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-21 07:57:56.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"b800f183fc509775f12ac7dbc328923b67ae8ef8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:58:06.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1551/","accounts_in_message":[],"_revision_number":13},{"id":"2e31e754a317fbf231986493c1f071d3a1045354","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 07:58:08.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914230/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914230/13/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"396c95f55a9db6898cac2692769b4abb5b3e8e72","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-21 08:00:19.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"2810ed5e9ba9ba69e3275b5310b926914d4028f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 09:32:13.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d34fcf9e9c97495899653d01bf2f7d3b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df4ca885742d433a8558608e6d65440b : TIMED_OUT in 1h 12m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84df5eeb7b0b4e9194277d9423720425 : TIMED_OUT in 1h 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cc6e59fc426dd31a717a8c3a74b260c11b4e2d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 11:16:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e64a466a385f4730af6c471a622414e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53323fbec92457ab7c099d64aa3e11c : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e6b778cfabf41788176431a39c5c6d9 : TIMED_OUT in 1h 28m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d50caa7f6aa44835bd5cf3b9bdae2ec9 : SUCCESS in 19m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049c595b97c04e47b07ce9e9bcf21db2 : SUCCESS in 23m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/217f360fbbfa4d8e8d23847b49a31afd : SUCCESS in 1h 12m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/08ba15652eee4854a2348c6c2b9f753d : SUCCESS in 1h 27m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/203f73dea8c7427c9f18c0c5185ca2ef : SUCCESS in 1h 46m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65a4655b50d6422a802a5075fadfa53f : SUCCESS in 1h 13m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e209d39cead04c0a94b77d890a4e4aa0 : SUCCESS in 17m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ac3233f6d1544c8ba519c3f36d1d3b04 : TIMED_OUT in 45m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a99516d302b44726946968e7ac290b89 : SUCCESS in 14m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5f91bad20584787b934a2c5b647a835 : SUCCESS in 15m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/729601adfaff49119a6b20a02279c803 : SUCCESS in 6m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/49122308e8b04646ad00ddec0d87b70f : SUCCESS in 8m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/addeb1d2309f42f0b2045f3bde188dba : FAILURE in 36m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab96dfd6c14342dabaffac4d535bbfb6 : SUCCESS in 8m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e34d0df00fb4fbcadc173c2f8ed4f56 : SUCCESS in 1h 50m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d2836ec1ea71448eb092f899efc666fd : SUCCESS in 3h 05m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/934072b3c5924b8d97f6b8af5d4cf041 : FAILURE in 16m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ea78dbe155ae4fa99a73b908d3076106 : SUCCESS in 2h 15m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1dbc8309418243bfa6f5116006196c34 : SUCCESS in 29m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f94c78c4ab744f448361805735737739 : SUCCESS in 39m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4bfc92f7778140ad8a2d34604976b3d0 : SUCCESS in 1h 52m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/27cda3b85d7449d59c0f6441d52a8ff8 : SUCCESS in 1h 01m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a1240df3507749b687d3b8f9bc979969 : SUCCESS in 1h 45m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0a26c1c6d0824c7d944476f5cefd7c80 : SUCCESS in 1h 52m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d8fa181043af453395fad4b45661fffa : SUCCESS in 46m 59s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb24652b663c46049ed6fd425fe3fb00 : SUCCESS in 2h 01m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/042cf2d394f9437080607a919d54ac9f : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":13},{"id":"680149876acf35b4830faf017726fd787af273c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 11:47:15.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/489/.","accounts_in_message":[],"_revision_number":13},{"id":"22de7138b0e68bb58d89e2e0ccdafdaf7c8e1a6e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-21 13:20:29.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1200e43352934bfdb7070c6d8df1d3d4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/975190a4dccc44fbacf72d78dbe6e8c4 : SUCCESS in 49m 22s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33fba219569f461eba2befd2d8efbbe4 : FAILURE in 44m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2c741d6bda8b41119bb6e28787b7a81c : FAILURE in 1h 52m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/805183e847a6462b90aba0c16c6a123f : SUCCESS in 41m 42s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/430a7a02c6304eecbe23fd836da9f594 : RETRY_LIMIT in 44s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/417d9b71dfe34a6f90fac6b129bfa328 : NODE_FAILURE Node request 200-0000033893 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ebe25fa750e4fcf8f0da665fceeeeb1 : RETRY_LIMIT in 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0181c021e87a401d90ea108b364818c6 : RETRY_LIMIT in 42s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"051ae7bf0fc7e39d96d010ffd3ebdeba606b5322","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 16:05:48.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/557/.","accounts_in_message":[],"_revision_number":13},{"id":"2ab399fbd00921bb8266e0745509b7428c00fde5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-21 16:40:19.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 33m 29s","accounts_in_message":[],"_revision_number":13},{"id":"cfd7bdc5fb6e50261a3658046e5fc7cd250319d2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-21 16:43:06.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914230 : FAILURE in 34m 05s","accounts_in_message":[],"_revision_number":13},{"id":"d942aba70859f96c13a0606f00b3c5a803a420a3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-21 19:52:04.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-13/FC/ : FAILURE in 9h 43m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-13/iSCSI/ : UNSTABLE in 2h 10m 9s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"4cec6328968ca1bc6402fd89ac517f7ce20f8fb3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-22 04:54:36.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"b8474dd0227b3226da5fcff10335d40d319acd4b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-22 05:11:19.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/afca271d04904451a0d8b4d41fff6802 : SUCCESS in 1h 12m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8dac1d1a0c8b40b6ae5100c1bbbe3147 : RETRY_LIMIT in 2h 25m 16s","accounts_in_message":[],"_revision_number":13},{"id":"951db5211379df47cf5f037c9f43fd3ce6870c2b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-03-23 06:34:16.000000000","message":"Patch Set 13:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/30/914230/13/check/fujitsu-eternus-dx-fc/fc6b001 : SUCCESS in 2h 30m 36s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/30/914230/13/check/fujitsu-eternus-dx-iscsi/42804a6 : TIMED_OUT in 3h 36m 42s","accounts_in_message":[],"_revision_number":13},{"id":"91e55638031cfa052b8881db4fa83344d37e94a8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-25 21:56:23.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/13/20250326030027/NEC-V-ISCSI : SUCCESS in 3h 40m 37s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/13/20250326030027/NEC-V-FC : SUCCESS in 3h 55m 53s","accounts_in_message":[],"_revision_number":13},{"id":"8d37cccc0313c3a53be817b2ae8753af51ed5dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"date":"2025-03-26 04:10:34.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"24df70afcd10a07bac4b9c66e2afafe319396d50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 04:10:44.000000000","message":"Patch Set 14:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1582/","accounts_in_message":[],"_revision_number":14},{"id":"4247a36eba5d8e4f8c614ab0b6cd433e09f96a72","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 04:10:46.000000000","message":"Patch Set 14:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914230/14/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914230/14/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":14},{"id":"ba4fea7b3b628197cf7272b362327c9df8677a4f","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 04:14:12.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"151724fd9c8deed002578e63d0aa959f4ac17b80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 05:18:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e00917704ad406186a4170db61c2aa1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/886a38609302410d9446e6d5b3f349f3 : FAILURE in 58m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dca48c82f8a94de88307938ea29767ed : SUCCESS in 12m 05s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c78eaefd9fe70933e86aa4b08551ed28507d8117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 06:53:31.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25585a8f504d4863b14117c7fc515f02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6fdac564b34d7388d31e55180392b1 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8e285a401494ed68288947effc36e87 : SUCCESS in 14m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9aba659cafe4440aa2708f349ada1464 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff24d1a2aa934ef7adda18b4a981e8b9 : SUCCESS in 16m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/a99b0295a10341fca950b410be08aee5 : SUCCESS in 59m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f20faeb0f2544bcebdc3b877b7df82fd : SUCCESS in 59m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/124dd5e2626149feace05db62fa7c486 : SUCCESS in 1h 21m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acf3ff0f81d8466c87dfb2f67d917351 : SUCCESS in 1h 01m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a1f6521f8ae464e8ecb24ef2d96e30c : SUCCESS in 19m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/569416e82b7a40448e61379885f9a29a : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b1bb08d69bed441c9d5ae572f18a86dd : SUCCESS in 8m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/15473a0fb7b24c5e84db0fc32aff67b7 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78d6d8f18acb4fadba731edc222e9bc2 : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a406698f464c45889ed6dc5bc79e1ff9 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/85521997e84744b8a8f098d131475104 : FAILURE in 19m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1c28c8c6eaf643998d28f5a5043c66e3 : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4e389d89df624766866cd863cd61f4ab : SUCCESS in 1h 57m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3cc7967664e54b8884da9cf487b97dd5 : FAILURE in 2h 20m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6e0c7ff5e9c74b6eb24d7c3364db4927 : SUCCESS in 1h 38m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b806bd01645949c99a84619ffbcb8776 : SUCCESS in 1h 57m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9a8a207f958341a08f08e604992f4b53 : SUCCESS in 29m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bfd669c178ff400bab5270e85bd6fcce : SUCCESS in 1h 01m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91fb753f89ab48a3811a905f0ac337e2 : SUCCESS in 1h 20m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e90f6d727c94994b6e772f54eae5864 : SUCCESS in 37m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc7e18ad40004e16adde3fab6c4c4993 : SUCCESS in 1h 30m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cb95aab08f6543b99028b99c96cd743c : SUCCESS in 1h 30m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3ec2dd23c064dd4afcb727eb9ae6cba : SUCCESS in 1h 38m 35s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7258b518a0f4abe97119f4e750c96e7 : SUCCESS in 1h 20m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05a7fe0672cb4339b7e351be49f67052 : SUCCESS in 47m 25s","accounts_in_message":[],"_revision_number":14},{"id":"7aa5076fb95aba4b051f87d01ffb3e27bc869877","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-26 07:00:06.000000000","message":"Patch Set 14:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-14/FC/ : UNSTABLE in 15m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-30-914230-14/iSCSI/ : FAILURE in 2h 15m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":14},{"id":"eeb89675dad923460031c0a5008636e1d0287e3a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-26 10:07:22.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"b580b9088e8a046e75400b78aeda4a96946bedb7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-26 10:24:12.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/668fba8844b74a0c828e38439fb00046\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5e1cf517cf724a81a18e2824d211b545 : SUCCESS in 44m 23s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/09c7847dc6604dbbbd2ee19722a5f9a2 : SUCCESS in 43m 48s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b6bbca724f64e5788905355e06e0072 : FAILURE in 1h 39m 20s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/356ea3d2d0294236b5d2d44ea237b04b : SUCCESS in 45m 31s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9cad70b90737472da7273d54497028c3 : RETRY_LIMIT in 46s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dfe49b820db04aed86cadbf8f3fc1deb : NODE_FAILURE Node request 200-0000035718 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/864f20ca564d43dc9d7779975c408d13 : RETRY_LIMIT in 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6670e6cd7d7d4c2bb8bb6fc6b1aee42d : RETRY_LIMIT in 43s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"832c1afeef3aa8d8b21600d471661d54f3c69710","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 14:21:49.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914230 : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":14},{"id":"93e31ec023f19a29fd410375c8cd0611d3c92612","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 14:40:24.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914230 : FAILURE in 34m 07s","accounts_in_message":[],"_revision_number":14},{"id":"f071f88298ad6067ad8ec41828a4b5e4f9c00001","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-26 19:07:46.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"cf3a8c63c31dcb48a0e057453a13436488e920e1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-27 01:31:48.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/585/.","accounts_in_message":[],"_revision_number":14},{"id":"d96b6913bb79c9c6410c4ef350d3bbb128973cc1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-27 02:06:36.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8e2fa18f2dd3450d8ce26b96682c1eab : SUCCESS in 1h 07m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/02a36ed38c9e47abb5b05dc863ff5966 : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":14},{"id":"401d52d8fe1646b075461118e2fd065826da03c1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-01 05:07:15.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/14/20250401095611/NEC-V-ISCSI : SUCCESS in 4h 07m 29s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/14/20250401095611/NEC-V-FC : SUCCESS in 4h 11m 01s","accounts_in_message":[],"_revision_number":14},{"id":"be95a268a6086a6aaec1bd95a5394d1be7faf77f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-02 13:23:16.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/14/20250402162712/NEC-ISCSI : SUCCESS in 5h 26m 51s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/30/914230/14/20250402162712/NEC-FC : SUCCESS in 5h 56m 01s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"72169222e02386fe32a0a20a44e12ec95be053ac","revisions":{"37f9898b18e2df97c57c530794c50c9dbca4d90d":{"kind":"REWORK","_number":1,"created":"2024-03-29 02:44:31.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/1"}}},"commit":{"parents":[{"commit":"821510c9f484ff5edf5f01b6701b77e88c73f842","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/821510c9f484ff5edf5f01b6701b77e88c73f842"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37f9898b18e2df97c57c530794c50c9dbca4d90d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37f9898b18e2df97c57c530794c50c9dbca4d90d"}]},"branch":"refs/heads/master"},"5f1f88e44da13907a6da1393b77f2de46be2b982":{"kind":"REWORK","_number":2,"created":"2024-05-21 07:06:06.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/2"}}},"commit":{"parents":[{"commit":"3778ce94b3a232a8736054c4dbd79eda77494dae","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3778ce94b3a232a8736054c4dbd79eda77494dae"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-21 07:05:37.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f1f88e44da13907a6da1393b77f2de46be2b982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f1f88e44da13907a6da1393b77f2de46be2b982"}]},"branch":"refs/heads/master"},"7f24117e1c4da431808180b2e5c02ace250d24c1":{"kind":"REWORK","_number":3,"created":"2024-05-21 08:55:27.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/3"}}},"commit":{"parents":[{"commit":"3778ce94b3a232a8736054c4dbd79eda77494dae","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3778ce94b3a232a8736054c4dbd79eda77494dae"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-21 08:55:00.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f24117e1c4da431808180b2e5c02ace250d24c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f24117e1c4da431808180b2e5c02ace250d24c1"}]},"branch":"refs/heads/master"},"420f85725f20f50a4613a44270b1384ed75a9c01":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-24 10:36:59.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/4"}}},"commit":{"parents":[{"commit":"311cb88fc4538426294df89fd800c5c0c5ba5f9f","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/311cb88fc4538426294df89fd800c5c0c5ba5f9f"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-24 10:35:25.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/420f85725f20f50a4613a44270b1384ed75a9c01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/420f85725f20f50a4613a44270b1384ed75a9c01"}]},"branch":"refs/heads/master"},"67cf1994b5046c559994306a7570cff7d7ed9abf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-27 10:12:26.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/5"}}},"commit":{"parents":[{"commit":"9ed8eade5dddc46468dea52af447fd34cd7c4e6b","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ed8eade5dddc46468dea52af447fd34cd7c4e6b"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-27 10:12:04.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67cf1994b5046c559994306a7570cff7d7ed9abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67cf1994b5046c559994306a7570cff7d7ed9abf"}]},"branch":"refs/heads/master"},"446ca62d89e0fb4217b6e1fd41a231ececb77971":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-28 03:48:44.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/6"}}},"commit":{"parents":[{"commit":"17c7098acdb14c5cbc8f907b424274c1c30a33d0","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/17c7098acdb14c5cbc8f907b424274c1c30a33d0"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-05-28 03:47:58.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/446ca62d89e0fb4217b6e1fd41a231ececb77971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/446ca62d89e0fb4217b6e1fd41a231ececb77971"}]},"branch":"refs/heads/master"},"64d75380728f8ab01ead977ae78bb2c21d8c9fc6":{"kind":"REWORK","_number":7,"created":"2024-06-14 09:11:22.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/7"}}},"commit":{"parents":[{"commit":"b1479c77799c8f4d4049da59ac048630e10cb30c","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1479c77799c8f4d4049da59ac048630e10cb30c"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-06-14 09:10:51.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64d75380728f8ab01ead977ae78bb2c21d8c9fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64d75380728f8ab01ead977ae78bb2c21d8c9fc6"}]},"branch":"refs/heads/master"},"95782c13668c185ba27b841226498bc724aea0f4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-07-12 08:51:42.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/8"}}},"commit":{"parents":[{"commit":"880336f40f8c46a830e34dc53ef68f014ed7b30b","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/880336f40f8c46a830e34dc53ef68f014ed7b30b"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-07-12 08:50:22.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/95782c13668c185ba27b841226498bc724aea0f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/95782c13668c185ba27b841226498bc724aea0f4"}]},"branch":"refs/heads/master"},"e0f3463489f7b5c910d77064718fb6025cb86469":{"kind":"NO_CHANGE","_number":9,"created":"2024-07-12 09:53:59.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/9"}}},"commit":{"parents":[{"commit":"7532b797ebe4597ddd5fe196c72fd9d8641e3d81","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7532b797ebe4597ddd5fe196c72fd9d8641e3d81"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-07-12 09:53:13.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0f3463489f7b5c910d77064718fb6025cb86469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0f3463489f7b5c910d77064718fb6025cb86469"}]},"branch":"refs/heads/master"},"9d524df5b28ca6969bec9c7c9abb57424fde68c1":{"kind":"REWORK","_number":10,"created":"2024-08-09 01:01:14.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/10"}}},"commit":{"parents":[{"commit":"2c778e17fc14c46b4d0374c55c3e7c96b531207e","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c778e17fc14c46b4d0374c55c3e7c96b531207e"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-08-08 10:52:48.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d524df5b28ca6969bec9c7c9abb57424fde68c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d524df5b28ca6969bec9c7c9abb57424fde68c1"}]},"branch":"refs/heads/master"},"65ec16d2756493870f920f01f7e7505ae2834740":{"kind":"REWORK","_number":11,"created":"2024-09-06 07:21:45.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/11"}}},"commit":{"parents":[{"commit":"1f0e33ad9827b7beea4171eb4d3a34d21c255381","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f0e33ad9827b7beea4171eb4d3a34d21c255381"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-09-06 07:15:08.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65ec16d2756493870f920f01f7e7505ae2834740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65ec16d2756493870f920f01f7e7505ae2834740"}]},"branch":"refs/heads/master"},"e7a3bda832330771d7e0d233d520d20e84c25ff9":{"kind":"REWORK","_number":12,"created":"2025-01-07 06:35:52.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/30/914230/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/12"}}},"commit":{"parents":[{"commit":"b476ee36e2e3c3b243313574386486f4b27f457f","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b476ee36e2e3c3b243313574386486f4b27f457f"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2025-01-07 06:35:35.000000000","tz":480},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e7a3bda832330771d7e0d233d520d20e84c25ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e7a3bda832330771d7e0d233d520d20e84c25ff9"}]},"branch":"refs/heads/master"},"ef0d56ecd9d516b81e587bfdd13bcde03e07657b":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-03-21 07:57:56.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/30/914230/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/13"}}},"commit":{"parents":[{"commit":"befbe7220346a7b90952b9166a0f6467b93c2cac","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/befbe7220346a7b90952b9166a0f6467b93c2cac"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-21 07:57:02.000000000","tz":-240},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef0d56ecd9d516b81e587bfdd13bcde03e07657b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef0d56ecd9d516b81e587bfdd13bcde03e07657b"}]},"branch":"refs/heads/master"},"72169222e02386fe32a0a20a44e12ec95be053ac":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-03-26 04:10:34.000000000","uploader":{"_account_id":37834,"name":"Masahiko Eba","display_name":"Eba","email":"eba.masahiko@fujitsu.com","username":"m-eba"},"ref":"refs/changes/30/914230/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/914230/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/914230/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/914230/14"}}},"commit":{"parents":[{"commit":"3e0439f7d0c2b593ab60b0151b9d263829cbbbae","subject":"Fujitsu Driver: Support volume replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e0439f7d0c2b593ab60b0151b9d263829cbbbae"}]}],"author":{"name":"xuq.fnstxz","email":"xuq.fnstxz@fujitsu.com","date":"2024-03-29 02:37:49.000000000","tz":480},"committer":{"name":"Masahiko Eba","email":"eba.masahiko@fujitsu.com","date":"2025-03-26 04:10:20.000000000","tz":-240},"subject":"Fujitsu Driver: Add support for data protection","message":"Fujitsu Driver: Add support for data protection\n\nData protection function provides cross-backend local copy,\nreplication or backup using Fujitsu Storage ETERNUS SF Advanced Copy\nManager Copy Control Module.\n\nDiffers from volume replication, volume cloned to remote site using\nAdvancedCopy Manager Cotrol Module appears in cinder\u0027s database,\nallowing for direct attaching or deleting the volume.\n\nChange-Id: I940049a9d58252980a8be47bdc93924e91436238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72169222e02386fe32a0a20a44e12ec95be053ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72169222e02386fe32a0a20a44e12ec95be053ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
