)]}'
{"id":"openstack%2Fcinder~914777","triplet_id":"openstack%2Fcinder~master~I9bce4bbeae78d8e4f95a5cd9de856edd54b05a76","project":"openstack/cinder","branch":"master","topic":"bug/2059821","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-01 03:10:25.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9bce4bbeae78d8e4f95a5cd9de856edd54b05a76","subject":"Replace deprecated glanceclient exceptions","status":"NEW","created":"2024-03-30 03:06:31.000000000","updated":"2025-09-06 16:27:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d7a91c82bdd357e2207dfe9d7056a0c3e1ba1cc0","_number":914777,"virtual_id_number":914777,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-06 15:24:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-30 03:16:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-03-30 03:28:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-30 04:10:15.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-30 04:11:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-30 05:21:26.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-03-30 05:25:42.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-30 05:40:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-30 06:18:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-30 07:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-30 10:15:21.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-30 13:45:07.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-03-30 13:59:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-30 15:38:17.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-03-30 19:39:37.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-30 21:54:42.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-03-31 01:07:09.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-31 08:03:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-28 16:29:04.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-09-06 12:34:29.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-06 13:12:34.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-06 15:38:18.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-06 16:12:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"755de0509e51cbb77b4581046e6bf4d6c562867a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-30 03:06:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"995ba7adebde6bd9cca61739c5ef2b8dda4aeff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-30 03:11:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0e4938379be3c2eec813d6dcd587177a030b0663","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-30 03:14:38.000000000","message":"Topic set to bug/2059821","accounts_in_message":[],"_revision_number":2},{"id":"d943653e1dc2f5bbbc6b22e7f96eb769d449bd2f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-30 03:16:52.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/319c0b7c7ade4aa38c513c85d960ac07\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c460e681e69140e7b4533fcc5a219085 : NODE_FAILURE Node request 200-0000002907 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"cf2a1058ba50e019a59951575200a386ecf48f61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 03:28:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16cbaf5dd2d4e988bf14b5a69081138\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3df9b3593731449997a7a5d4a4a3f666 : FAILURE in 10m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d496a34dbee6473a9d9545efc2d5a9ea : FAILURE in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db26eb0c4c70189ca16c2a85f716886a2488d2ca","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-30 04:10:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : FAILURE in 55m 10s","accounts_in_message":[],"_revision_number":2},{"id":"0109215139e39acccbb4ee9a8c3c4caa9f900eb2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-30 04:11:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914777 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":2},{"id":"bdf71f3f24deb7c57b516d9c37f96d2a25d1a25d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-30 05:09:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : FAILURE in 55m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8bdf00b07e8ee85d7665ec8aa74598128263d750","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-30 05:21:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/149/.","accounts_in_message":[],"_revision_number":2},{"id":"81bdc5d682ff42eb0491629099eb3a5c6c9b21b2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-30 05:25:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/77/914777/2/check/DellEMC_PowerFlex_3.6/b03f69f/DellEMC_PowerFlex_3.6/1110 : SUCCESS in 1h 32m 27s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/77/914777/2/check/DellEMC_PowerFlex_4.5/4742bd7/DellEMC_PowerFlex_4.5/87 : FAILURE in 2h 14m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c11731f7e9ab30960c90af635d23ef9528554546","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-30 05:40:36.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/77/914777/2/check/DellEMC_Unity_FC/6caf5f7/DellEMC_Unity_FC/1604 : SUCCESS in 1h 19m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/2/check/DellEMC_Unity_iSCSI/f489ac2/DellEMC_Unity_iSCSI/1635 : SUCCESS in 1h 28m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"82bef1ba5eb1db60fdfb8c934b993dc6b227008c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 06:18:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4836568722234955b7f8474569c4c1d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79bd690457fb4367bf0cff3aa852170d : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cd4a02fe5f04536b82909163124e6fe : FAILURE in 11m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31731b2d7f704ea991922e454541899d : FAILURE in 10m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdfc6f19daa847c5817dbd88ec5d4fae : FAILURE in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329e2fdf530d4d48930b656bac55d533 : SUCCESS in 11m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5cd554e20e0d48ae81c9171179e183e0 : SUCCESS in 52m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c45ed13cddf44c68394a996b18164f7 : SUCCESS in 1h 08m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdbdfe0cd2a9420b9d8a0933535d7368 : SUCCESS in 54m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1e2db61360e244b9b242ab264b1a1a60 : FAILURE in 12m 52s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/75ce428a616d4681a8064dde90a3e759 : FAILURE in 11m 29s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9e38b056a463448ba06eca444a6c6c23 : SUCCESS in 7m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8b8a494d13e48f299536588ace2230b : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3dac4bbdf4754c28a36d958aa5515f73 : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/396c5224821a43c0abd211934622d200 : SUCCESS in 8m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8322d6d9506742bab07580dce7d12d8f : SUCCESS in 38m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9c760148825e48aaaf28d3606d35621c : SUCCESS in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/976f93f6c8384da99c1a5eb75bb7fb87 : SUCCESS in 1h 52m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8e63aadadf24e8785460dc399b15373 : TIMED_OUT in 3h 03m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/35f6898b44ee42f184cbd97d455c3c92 : SUCCESS in 1h 46m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/09bd6043247a4880a94d7c204e71d9da : SUCCESS in 1h 33m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e2fdd1c888984579b45ae14f19014ba4 : SUCCESS in 27m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8ab0691b66f747cdadab084bf8ee8936 : SUCCESS in 1h 11m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0ce6ef5072e6446198585398a08166d7 : SUCCESS in 1h 33m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e7bb0c488e8646539ba30b470319e63c : SUCCESS in 52m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/070651b728154b2883808bbd373ab99d : SUCCESS in 59m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3bad097c1703478884f3f43b3f8e0164 : SUCCESS in 1h 12m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7bcac55670f34ca1a289b104087d6ec2 : SUCCESS in 1h 06m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58cdfd6b4a134c01a7246f574e17fe49 : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":2},{"id":"49bffa142f54127d965062012ef805ddde26f5d0","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-30 07:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/914777/1/fc : SUCCESS in 0h 46m 18s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/914777/1/iscsi : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d3e84ccca45d0e18d937d799e4c4bf77cfcb88a1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-30 07:00:10.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/914777/2/fc : SUCCESS in 0h 48m 36s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/914777/2/iscsi : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":2},{"id":"e66f506b957e2bff7e3e3764e8f592ad200f6f2c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-30 10:15:21.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-1/FC/ : FAILURE in 7h 8m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-1/iSCSI/ : FAILURE in 4h 1m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"9647ad59dc6a8460606be199c1754df1e49cbc4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-30 11:05:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6e0714b6d24edfab6c6331e09a17db0c25b8e31f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-30 11:11:52.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5c7ea9f0c62b4c4a95d4c7c10c8b25cc\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c7807a34aad342158713c831cc993796 : NODE_FAILURE Node request 200-0000002908 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"295a465fd60bf8cc8cab125919c34da0f9c2fd53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 11:16:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1bf313b43214e7da2f6b26934eb2c7b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00e604a649fa4e6abb1e491ba9f0e6d3 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b790c8da3a0c42f8a3d570ea98cded01 : SUCCESS in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0785a5c937a94595e1fc6215c4d8c408828915c4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-30 11:35:23.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-2/FC/ : FAILURE in 7h 58m 30s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-2/iSCSI/ : SUCCESS in 2h 34m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"a7804f33959c76d2cf7ce0a9054565623f68c882","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-30 12:01:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : FAILURE in 54m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8bf994722b2303f7f319fa49cae278ddf1adb240","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-30 12:06:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914777 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":3},{"id":"4c5501647910ee771cfaab9bdddefc12a6f994d6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-30 12:35:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/150/.","accounts_in_message":[],"_revision_number":3},{"id":"9d3cf139eacd1f8e25bfb21fc7bb925dcce45640","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-30 12:37:48.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_Unity_FC/a52f45b/DellEMC_Unity_FC/1605 : SUCCESS in 1h 19m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_Unity_iSCSI/91bf965/DellEMC_Unity_iSCSI/1636 : SUCCESS in 1h 32m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d0e8264802054d67a5beace92b68605594e472d1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-30 13:06:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : FAILURE in 1h 01m 12s","accounts_in_message":[],"_revision_number":3},{"id":"0f617d8969d201afd4c68fab4bb44bc7f6d52118","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-30 13:35:16.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerFlex_3.6/82b1bdf/DellEMC_PowerFlex_3.6/1111 : SUCCESS in 1h 37m 41s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerFlex_4.5/e635567/DellEMC_PowerFlex_4.5/88 : SUCCESS in 2h 29m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"629c4701ac274657bb84fa667efa6e7e49f11967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 13:38:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ef3e2d7a3e460187faca259b876191\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af8ba57fffa49e98a07a4065775e846 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7b3017db3f346668ee0925707f926ee : SUCCESS in 9m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d885e7b6d0b04837ba88c11b8160480f : SUCCESS in 10m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b90bfec85374e569976252d02e01913 : FAILURE in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a4937ac96b459cb87ffa01af09862c : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/81121bbb82bd410da17a87ee3c763f9c : SUCCESS in 1h 04m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68da32d830d24f6d82873d2358c9339f : SUCCESS in 1h 17m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13b54712187943bf851b0aa84188e133 : SUCCESS in 51m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/98af458f2cd94395b1f096d8568b9e2a : SUCCESS in 12m 43s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d250bfb73dd54e78b88900c0bd08a3c7 : SUCCESS in 13m 41s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fe74da1d06bf4f8c93f572ea8e4f384a : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c6e73b74a884512b6fdd7f19b521978 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d32e46ebf1574e3586c70ebf33d78551 : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/516b44d3d7b54775b7abddff300106e2 : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/812fe99b93c14ba08902524d5993876d : SUCCESS in 28m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d591d1e682f84f2c9f4a73fa4320a2ab : SUCCESS in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b1583f234542499ab2de77d959dc4f88 : SUCCESS in 1h 24m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c8895f1fb64140c39bc64563242eb458 : FAILURE in 1h 53m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d61027f109dc4668820c849ab53d826e : SUCCESS in 1h 08m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6f70e9d1a478484e82b745888155ffc7 : SUCCESS in 1h 22m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/92d68a3437db4977b56c6557c3ff5587 : SUCCESS in 21m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a735c506695d4388b3f8389aedb2fb0f : SUCCESS in 1h 06m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0aa90ecf9579499993d04a3099ece8ff : FAILURE in 1h 40m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cc1371e7fa5545d285d352da172f07f4 : SUCCESS in 27m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bef056728c3640979460eb217a9e4a58 : SUCCESS in 1h 24m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/00433ad78eda40be866e6c45f66199f8 : POST_FAILURE in 2h 26m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/752fb011d85b44028de24f88dd6bcff5 : SUCCESS in 1h 33m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d874ac563679447281a82251c612e5c8 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":3},{"id":"81493bbecd8fa2c44ae9f88bf9ea00c83b2105ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-30 13:45:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_03_2024_13_38_57\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"52131adae1c6c058f88770454757788898de528a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-30 13:59:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/914777/3/upstream-check/cinder-tempest-plugin-ontap-fcp/8f581f8/ : RETRY_LIMIT in 3m 59s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/914777/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/ae9022d/ : RETRY_LIMIT in 3m 34s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/914777/3/upstream-check/cinder-tempest-plugin-ontap-nfs/c45d4bf/ : RETRY_LIMIT in 3m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/914777/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/b70ce43/ : RETRY_LIMIT in 3m 47s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b98cde208e87c6a75dafc8f575646aa07fbe64ce","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-30 14:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/914777/3/fc : SUCCESS in 0h 41m 04s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/914777/3/iscsi : SUCCESS in 0h 57m 59s","accounts_in_message":[],"_revision_number":3},{"id":"989e8c89fe829dc603e1c6fe6ef46326b351c196","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-30 15:30:26.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_Unity_FC/612cab3/DellEMC_Unity_FC/1606 : SUCCESS in 1h 22m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_Unity_iSCSI/5d21263/DellEMC_Unity_iSCSI/1637 : SUCCESS in 1h 51m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"0275f61a04530d86d376cb96bde779f60c837f20","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-30 15:38:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerStore_iSCSI/418a9b9/DellEMC_PowerStore_iSCSI/1213 : SUCCESS in 1h 11m 40s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerStore_FC/d7d5f4f/DellEMC_PowerStore_FC/1384 : SUCCESS in 59m 02s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerStore_NFS/aace212/DellEMC_PowerStore_NFS/903 : SUCCESS in 1h 05m 32s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerStore_NVMeTCP/0ef5898/DellEMC_PowerStore_NVMeTCP/764 : SUCCESS in 1h 59m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"832f30695565ab4f9fdde9d6dbd9832a737c07ac","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-30 15:57:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerFlex_3.6/141c347/DellEMC_PowerFlex_3.6/1112 : SUCCESS in 1h 47m 45s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerFlex_4.5/2504121/DellEMC_PowerFlex_4.5/89 : FAILURE in 2h 18m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"fd15426edda529425744bcbd3f4e80ea31d4eb53","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-30 18:30:21.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-3/FC/ : FAILURE in 7h 24m 46s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-3/iSCSI/ : SUCCESS in 4h 7m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"3ddf21d7bd8134dbb1adb287bb98b46dec92e767","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-30 19:39:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerMAX_FC/1ccca15/DellEMC_PowerMAX_FC/369 : SUCCESS in 4h 55m 55s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/3/check/DellEMC_PowerMAX_iSCSI/b46d296/DellEMC_PowerMAX_iSCSI/390 : FAILURE in 6h 00m 32s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"b9f1d3670cb390c6972fa94ffa2732728d1aa755","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-03-30 21:54:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/914777-3/HPMSA-iSCSI/ : SUCCESS in 1:53:11\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/914777-3/HPMSA-FC/ : FAILURE in 2:05:36","accounts_in_message":[],"_revision_number":3},{"id":"63470205b18e0a783228011977a8acde1d0f4439","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-31 01:07:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/77/914777/3/20240331033148/NEC-V-ISCSI : SUCCESS in 3h 12m 25s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/77/914777/3/20240331033148/NEC-V-FC : SUCCESS in 6h 35m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d74303bbdc28080f749fbce0117ba5fd2f9b9d8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-31 05:01:57.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d06c21ed38da5947ab42ab3ce6bf4fa3fbe02741","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-31 05:08:46.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b882f5c85589420781c3a4ed9ece12f2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/835710b1d43745168bbc3e1da9dc20a5 : NODE_FAILURE Node request 200-0000002909 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"1775bee66420a2ae4f2e4a4399b80a413c31b0df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-31 05:11:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53662fa853e3496381f82a6b3d32e844\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6074b8130f0a4d909dc88185c8261a67 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d43f68417fa44ede8a7c6b7e90f963c0 : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7f5c9aacf0d2e49c79b167f4cef80539aa10605d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-31 05:58:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914777 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":4},{"id":"c709088e06591a8344444b38249f3422b3f438c1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-31 05:59:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : FAILURE in 55m 46s","accounts_in_message":[],"_revision_number":4},{"id":"682f9a35133d13ccc301c8cdfc6624ea00f63039","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-31 06:18:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/151/.","accounts_in_message":[],"_revision_number":4},{"id":"a6f3bd8e67ca2a436c7c9cf49f6f099d55cc1c0e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-31 06:48:07.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/77/914777/4/check/DellEMC_Unity_FC/2a11257/DellEMC_Unity_FC/1607 : SUCCESS in 1h 20m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/4/check/DellEMC_Unity_iSCSI/da20cac/DellEMC_Unity_iSCSI/1638 : SUCCESS in 1h 45m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"a4324cec3e5d65fb60528bf83c1a85e7e6e58f56","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-31 06:58:01.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : FAILURE in 55m 07s","accounts_in_message":[],"_revision_number":4},{"id":"e3051472b2c1b021ab1823e8f036346b27881343","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-31 07:11:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/77/914777/4/check/DellEMC_PowerFlex_3.6/42629a2/DellEMC_PowerFlex_3.6/1113 : SUCCESS in 1h 47m 02s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/77/914777/4/check/DellEMC_PowerFlex_4.5/ba694bf/DellEMC_PowerFlex_4.5/90 : FAILURE in 2h 09m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"6d3ffca2cdab6b6dc0b54a1009d64ff4f6ef6bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-31 07:17:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83d539922f4a4f77a9452ce6359e6668\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10ba577d6f7b4201a98640ca93076f46 : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c2e2610f0e245b0897f36f7c6616d4b : SUCCESS in 10m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffcf6ad35bd843daad20359a71610e8c : SUCCESS in 8m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1529a8011618411d904c151953af7e46 : FAILURE in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5c0f10c6384d359b974b6466c1f10c : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/a705d510d10940c493b3fa6a3a7b0b5b : SUCCESS in 38m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de635a23707b4a2f859353a2ebd435c4 : SUCCESS in 1h 08m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c62a143fcb0d4fd6a1f0e88fec47d108 : SUCCESS in 1h 00m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d73d695bb56d48b9b186774d40629aec : SUCCESS in 11m 23s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2ea45f072d3044eabd4ab69e7ecfd5da : SUCCESS in 12m 51s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dbc90bd881d0461fbed58f32a5d23dbe : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41d9b47ab3c94c47ab1224e6a2833af0 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42ed965f196e4d31a64bd84f6b55e2a6 : SUCCESS in 9m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92dd8b23c9af4d54882cd4f406199db7 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d936cdf5cee4b42bd526265ef13aeab : SUCCESS in 27m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff58bf205863423d9b13442124852baa : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/454fc9bd7331404ebb31357a0c04dcab : SUCCESS in 2h 03m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b5f8b4f95c424114b37f58e63c498475 : FAILURE in 1h 52m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5319d0d2bbbd49b597ff4295c89bf403 : FAILURE in 1h 44m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4cc1f57cdadf487d8755a5a19e9d4756 : SUCCESS in 1h 13m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d428d31270c4478db61513a3222f78ca : SUCCESS in 22m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/01eeb70f551a41b1a76f257148dc62b9 : SUCCESS in 47m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4993b7d2fa8045c294007ba459fc0782 : SUCCESS in 1h 25m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6542f37432774e37be81a39d60388d7f : SUCCESS in 46m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d71e1704d1cb44379212e2c3e794fe37 : SUCCESS in 1h 28m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1982d31805ac4c6bb9101980111aec3e : SUCCESS in 1h 48m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7823f3ff5ea7409a92a5159e158bc608 : SUCCESS in 1h 11m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8772c6ae3e24507befd83aa6cea4efe : SUCCESS in 44m 00s","accounts_in_message":[],"_revision_number":4},{"id":"6caad7d487949e0dd152264c44d87d1acac93d8f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-31 08:03:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/77/914777/3/20240331082449/NEC-ISCSI : SUCCESS in 4h 10m 33s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/77/914777/3/20240331082449/NEC-FC : SUCCESS in 8h 38m 35s","accounts_in_message":[],"_revision_number":3},{"id":"a657eddc74b93018c4d941590c5bccb5a4807cd2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-31 12:49:35.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-4/FC/ : FAILURE in 7h 47m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-4/iSCSI/ : SUCCESS in 3h 10m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"92badaeb05ce4c2158a34c065507157f0c18399b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-31 12:55:25.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/77/914777/4/20240331170331/NEC-ISCSI : SUCCESS in 4h 12m 12s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/77/914777/4/20240331170331/NEC-FC : SUCCESS in 4h 51m 50s","accounts_in_message":[],"_revision_number":4},{"id":"cd8ee517dd9487149974f32696c94d4eeee479d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-31 13:15:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e66c3b954e351e0f175daa93a7c0ceea8b400d40","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-31 13:23:27.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e43b92042e0143d5a4be2e622ad8d1a3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a25859b8a9504389aae5c862db93b33b : NODE_FAILURE Node request 200-0000002910 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"882bc504ce79a12eb7ea238ecee56256249cb286","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-31 13:26:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1015fb6b957e43669d6c7674140f39af\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4d82b6a84e941e081dbc0d8637e28a2 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/760452132846496d8e466776c1b1696d : SUCCESS in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1bb46c0d9554eb4016b7eec92e64f18188f8142d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-31 14:12:12.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : FAILURE in 54m 53s","accounts_in_message":[],"_revision_number":5},{"id":"d0fa595b14be19b6295edef2e32f5c28b3f08520","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-31 14:16:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914777 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":5},{"id":"3adde33dc0fcd52b271bf488fca4de2627799e39","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-31 14:32:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/152/.","accounts_in_message":[],"_revision_number":5},{"id":"151dc24b42cc326191a165e4926900bb8658958e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-31 14:47:59.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : FAILURE in 32m 19s","accounts_in_message":[],"_revision_number":5},{"id":"7a8b4493f46efbc2f454e576d4523be176463da2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-31 15:03:15.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_Unity_FC/c0a6c17/DellEMC_Unity_FC/1608 : SUCCESS in 1h 39m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_Unity_iSCSI/c4733a3/DellEMC_Unity_iSCSI/1639 : SUCCESS in 1h 47m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"02525b2b63ce914ef96787442104405305a1ac4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-31 15:24:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf6d2a1fe0bc4c93a49478227f08287d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9779cbf589849eb990957011348a2b4 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95f41ae886e64262b336cbde991442f2 : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/817b4fc3a7534750acb3f6b1a4dc8a36 : SUCCESS in 10m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d38bd3844bee4aa29504e25d7819832d : FAILURE in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7fdc47789240ec9fc65af25e0322db : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/697bcec3b8cc4388bb757737dcd50345 : SUCCESS in 43m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d240310e4ab4667a0a75db6aa772d62 : FAILURE in 53m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f414b1573feb492fad8a20ec96d7fe60 : SUCCESS in 37m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84fbb1412e7a480a999308be877b544b : SUCCESS in 12m 10s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b0dd9df0f79a4bf9af677da289538659 : SUCCESS in 9m 37s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c8ec7de20d6f458aba6d90b3f7d73bcf : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f1d41a7eb64c432b9892ed1ee1191b49 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d366555e3eb54307a8195ae0681df308 : SUCCESS in 6m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6aaff0dac07744a9b687748bd9c7588d : SUCCESS in 4m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91bdcca7e1d74a2297bdf8d57d71d05f : SUCCESS in 28m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b4a2602a8bf1446e902f7609d1c36754 : SUCCESS in 4m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2e0e854ca9645eb889234cf9db0ced2 : SUCCESS in 1h 23m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d75799d571647fb99c495b876ba16a0 : FAILURE in 1h 50m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/347ee0fd231441fd9b6960880c370568 : SUCCESS in 1h 07m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/38d83f19910d4d5e9646ba7645de03e0 : SUCCESS in 2h 03m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/667e14f36be1420a94d9a4cac1043be4 : SUCCESS in 38m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a37922707166402a8d5a4cc21e334af0 : SUCCESS in 1h 07m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8bf6006eedf4824b2ce9a4efe46aea6 : SUCCESS in 1h 29m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d2f209b24d7b414f97bd5b9fa64edf85 : SUCCESS in 55m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bab4138a820541ceab6eeff1529c63db : SUCCESS in 1h 29m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5783b7d5294442bdbfe6f2b6f32e1c76 : SUCCESS in 1h 37m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9eadf028873a42a8b341de7993b2db75 : SUCCESS in 1h 36m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/551d3b37b62d47879008f78c412c8b8a : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":5},{"id":"5ed677b1e49c841e34b0f54b0b73013f91416e39","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-31 15:35:32.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerFlex_3.6/2ba5f28/DellEMC_PowerFlex_3.6/1114 : SUCCESS in 1h 40m 27s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerFlex_4.5/21bcf1e/DellEMC_PowerFlex_4.5/91 : FAILURE in 2h 19m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"31eefce2bc38d37bb6958da087abe918a78410ce","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-31 16:00:09.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/914777/5/fc : SUCCESS in 0h 41m 29s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/77/914777/5/iscsi : SUCCESS in 0h 56m 43s","accounts_in_message":[],"_revision_number":5},{"id":"dbf876fdc6398dbe66a9c6771b4e100abfcac94f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-31 21:35:15.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-5/FC/ : FAILURE in 8h 19m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-914777-5/iSCSI/ : SUCCESS in 3h 22m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"d88bcd55bdb338115e683218525238839f5428ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-01 03:04:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6d5f8e293115ee5651af40284063227dda577530","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-01 03:10:25.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/be6d1bfc5cfb4df9b0d2e18ab5ea3410\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4a658076cfc24acc9ad8effc75377488 : NODE_FAILURE Node request 200-0000002911 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"96a7bf57bb8dfe1837c31a4c5735e2a97b1c28df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 04:06:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d66559c69f44dd39a68f979f51128e2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/170d3c9adcaa4cb09fe8faa341ea6b62 : TIMED_OUT in 1h 00m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe475b6c95184a20b8c0f478b9fd871f : SUCCESS in 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"32439a10291c2b60a81f93a6f5cefbf43e5ed5c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 05:04:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c0da17cafde458180e1075ab81fa0b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec9c84595cc4c00858129b4c918759e : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd2b90ab9ff6467da1872bedea0cbd41 : SUCCESS in 10m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbea373ba1cb40578329f56dad3fbb95 : SUCCESS in 10m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e7a6e799f0b41b2b7e303a2a393fef2 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c0aeaa3f0e4839bb257807e85e6fcb : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/3eaf47dd853d4aa4848d5f397e96a3e9 : SUCCESS in 48m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c286a950c934938b66c8b9fc70ea114 : SUCCESS in 1h 17m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15dc00b449f04638a85c37a0162985ba : SUCCESS in 48m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/588968953c674c898b281e0b9056565b : SUCCESS in 19m 24s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7c96d2103138420c929379ebe0e323ae : SUCCESS in 11m 51s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/169f397d42a544c5b36b0bfc84bc9bda : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41ae38da9df74a33b5d9548c7f115ebd : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9dd5b2a437d54ab896035cd71b2e0144 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc0f73a6fb724865ac67ef5b61743b02 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f0139c8946f047c28016cc0ffd03f158 : SUCCESS in 30m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b947266214942f5b8d05e501e7e755d : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/98b8f434a48d4eb3871e046599d0ac1a : SUCCESS in 1h 43m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/67e2014b8ffa44d9a560a2bc445b63a2 : FAILURE in 1h 50m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/47e28df00e08404fab41c52f1f9dff80 : SUCCESS in 1h 28m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/00b76a57aff940fdb121e286e0387aff : SUCCESS in 1h 41m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0439aea335964d0baf58c5c0e0ad46e8 : SUCCESS in 27m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/81d82d295c9e47d587a0615354eea367 : SUCCESS in 1h 13m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86c0396d167a4ffbb1e6b8f329883b2e : SUCCESS in 1h 11m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f5f67ec5aea41c28d9d4aa2fad95387 : SUCCESS in 33m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/11b508fd48bb46ccafc9602c115cfc8b : SUCCESS in 1h 26m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/597be6646a7746bf9a42af11664e17c1 : SUCCESS in 1h 40m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d00a0b45cc8343149408a2d17d944bfb : SUCCESS in 1h 12m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd4e749360dc43e691075797621bda1c : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":5},{"id":"39f5b8c1443e1c0212b1fd28310c853cc6ff7761","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-01 05:10:52.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_04_2024_05_04_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"5dd7554d80139b17bdbae967cd2a9bb0392f80b0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-01 05:30:40.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/914777/5/upstream-check/cinder-tempest-plugin-ontap-fcp/59a53ce/ : RETRY_LIMIT in 3m 42s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/914777/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/346a077/ : RETRY_LIMIT in 3m 50s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/914777/5/upstream-check/cinder-tempest-plugin-ontap-nfs/6bdd640/ : RETRY_LIMIT in 3m 43s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/914777/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/0dd9a6f/ : RETRY_LIMIT in 3m 26s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"7bd32445240a5aaf532742b73946a77ac65d1263","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-01 06:36:45.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_Unity_FC/b90bae7/DellEMC_Unity_FC/1609 : SUCCESS in 1h 18m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_Unity_iSCSI/34f5aa6/DellEMC_Unity_iSCSI/1640 : SUCCESS in 1h 31m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"3527d93dda7cda4bce6a196850327d13c29bfa1b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-01 06:37:23.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerStore_iSCSI/d7aeb91/DellEMC_PowerStore_iSCSI/1214 : SUCCESS in 1h 20m 55s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerStore_FC/f725b9c/DellEMC_PowerStore_FC/1385 : SUCCESS in 54m 07s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerStore_NFS/2c3dfa3/DellEMC_PowerStore_NFS/904 : SUCCESS in 1h 15m 58s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerStore_NVMeTCP/4fd2971/DellEMC_PowerStore_NVMeTCP/765 : SUCCESS in 1h 32m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"fbbbabc16a409de270f3708cc8dc8626a8d97e43","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-01 07:24:40.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerFlex_3.6/9d74841/DellEMC_PowerFlex_3.6/1115 : SUCCESS in 1h 39m 08s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerFlex_4.5/451ffe6/DellEMC_PowerFlex_4.5/92 : FAILURE in 2h 19m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"079eefb6cc7260ca90c71d3836f010c479129986","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-01 10:16:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/77/914777/5/20240401063225/NEC-ISCSI : SUCCESS in 3h 48m 02s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/77/914777/5/20240401063225/NEC-FC : SUCCESS in 12h 43m 52s","accounts_in_message":[],"_revision_number":5},{"id":"7f2248cb6036e7e55f17ace561a285bb91fd080d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-04-01 10:57:31.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerMAX_FC/0cc86ac/DellEMC_PowerMAX_FC/370 : SUCCESS in 4h 58m 25s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/77/914777/5/check/DellEMC_PowerMAX_iSCSI/5e53e06/DellEMC_PowerMAX_iSCSI/391 : FAILURE in 5h 52m 43s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"70dfb5d71d81c4d50c64281975cf84d286a8b67d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-04-01 23:50:37.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/914777-5/HPMSA-iSCSI/ : SUCCESS in 1:48:57\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/914777-5/HPMSA-FC/ : FAILURE in 5:33:04","accounts_in_message":[],"_revision_number":5},{"id":"2edce37c72ae57d5a278bc337d2951ac384ba4ac","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-04-28 16:29:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/914777-3/Lenovo-iSCSI/ : SUCCESS in 2:05:56\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/914777-3/Lenovo-FC/ : FAILURE in 3:33:07","accounts_in_message":[],"_revision_number":3},{"id":"216b165a7d49c05d2a0ad337b2aeffb185ec7197","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-04-29 13:13:12.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/914777-5/Lenovo-iSCSI/ : SUCCESS in 2:07:27\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/914777-5/Lenovo-FC/ : FAILURE in 3:34:03","accounts_in_message":[],"_revision_number":5},{"id":"54c2aad57b5cf922ed07c83cf144bfee7de4252d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-28 09:06:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/77/914777/5/20241028140646/NEC-V-ISCSI : SUCCESS in 3h 29m 52s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/77/914777/5/20241028140646/NEC-V-FC : SUCCESS in 3h 59m 53s","accounts_in_message":[],"_revision_number":5},{"id":"cc5bc1591388911a3f71481285537546657b751b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-06 12:33:02.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"20c445283b6c87a804e7223d4d236ad624317218","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-06 12:34:29.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"400172c8c814cf4392c9dd216bf568da84808102","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-06 13:05:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/117304bd4d924e548c7f51c2632354e8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9951d3e11d734e8cb54c54a99b6dd828 : SUCCESS in 25m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c271b78fb8094015b581497221f6a37b : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/96a24b08664a4bd8b785190d44dad244 : SUCCESS in 30m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2ada5c9d988cc90e26dada97260db2547402d5a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-06 13:12:34.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/486/","accounts_in_message":[],"_revision_number":6},{"id":"7f7b84e61c2df89a0c8c110a09122ebe88ab6e93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-06 13:32:01.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914777/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/914777/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"4553308cf0690f293e2469a9dbc9b754de192859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-06 15:24:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f03ea811fa1d4197a5c3563b94da87a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c27f30f685a4ee7a4711e640fdb22ac : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63115071e38048a6a8dc5d0e198fdd69 : SUCCESS in 13m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37b093f9e786441196c23dfaeed19f7b : SUCCESS in 12m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/115b5c54a8664317bca236982cb978e1 : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7fbcaa6f1d46aeb1201db9cdd37237 : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/1627588269c34f4eae9cf5d02b3bfa85 : SUCCESS in 1h 12m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4af7a085acf34fccbf05f315cda9cfaa : SUCCESS in 1h 05m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3be42210c5954086b83bbd8c5a7ea5ef : SUCCESS in 1h 42m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c670112a916843edb4fdce63d363f08d : SUCCESS in 1h 02m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/19bacca8be444217931227df79fe5abc : FAILURE in 21m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fdca46f91e3746da8e9f97b9e032416b : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6dcacd69358946969e77544950062536 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d80d84d639a44ab888cb4d786d86640e : SUCCESS in 5m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bed5bd0fd7eb4c17bf21361721ae3d2a : SUCCESS in 9m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5fd6f3d00c244b78953447ec700ccd05 : FAILURE in 35m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa89efe534ba4ecabc37fbebf5301b64 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d2fed922b67945f0906ec92bb166e0c5 : SUCCESS in 1h 55m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/33c914d98dac456990a86f14d422190a : FAILURE in 2h 44m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ff064faf4fd74cf3b75e3ba3a39794aa : SUCCESS in 1h 50m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b5822bb76e2b4d7cad204c52aa9679e1 : FAILURE in 17m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/52ed73b9bff84f1893feefa1cd03222e : SUCCESS in 30m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/231c7642fb6e436fb4d5cde15ac4338a : SUCCESS in 1h 02m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/53f6f236c556426d850693b5a144d915 : SUCCESS in 1h 44m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b9b07f0de1904f969c207d9105cdc349 : SUCCESS in 39m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5963a758b52846ac9cd0c7a05773000f : SUCCESS in 1h 36m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/195d2af11607485fa641ad2143f00c9f : FAILURE in 18m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54b04aed21604b439a674d2a764d7695 : SUCCESS in 1h 28m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fe6e648a7774263836dbd055e6e1166 : SUCCESS in 1h 26m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bf940fb665d4da18f12fb1768f2405a : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":6},{"id":"1c3901e4a170ba13d5d9910b88cbb0be2c99e07f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-06 15:38:18.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/20e70d52853b45bd8ffd8a96e09af37f : SUCCESS in 47m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f669aac7fd984ad6ba5d5589ae72e915 : FAILURE in 56m 24s","accounts_in_message":[],"_revision_number":6},{"id":"b631f83baa0ca7f0a7c512a4d5e0d8255f03916c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-06 15:39:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1707/.","accounts_in_message":[],"_revision_number":6},{"id":"08cf7512d8fc56575d102c0e07c8babafd265744","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-06 15:48:40.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1843/.","accounts_in_message":[],"_revision_number":6},{"id":"926e4edb950ae9c7b0a6325e8a68cefe86366c8f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-06 16:12:00.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":6},{"id":"d24006fbf0e45c4e72bc87aa8596a16c6b38ce25","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-06 16:13:20.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/914777 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":6},{"id":"d7a91c82bdd357e2207dfe9d7056a0c3e1ba1cc0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-06 16:27:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/914777 : SUCCESS in 1h 16m 02s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"193ac168d74d867b84afda0bc949197c5953f6a9","revisions":{"85e4beae93ce037633fafe11c362968c9fbe182a":{"kind":"REWORK","_number":1,"created":"2024-03-30 03:06:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/914777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/914777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/914777/1"}}},"commit":{"parents":[{"commit":"d94ecc22c036d060c0a52bd5130281e244a769e5","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d94ecc22c036d060c0a52bd5130281e244a769e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 03:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 03:05:33.000000000","tz":540},"subject":"Replace deprecated glanceclient exception","message":"Replace deprecated glanceclient exception\n\nExceptions without HTTP prefix were deprecated in glanceclient 0.4.0.\n\n[1] 354c98b087515dc4303a07d1ff0d9a9d7b4dd48b\n\nChange-Id: I9bce4bbeae78d8e4f95a5cd9de856edd54b05a76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e4beae93ce037633fafe11c362968c9fbe182a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e4beae93ce037633fafe11c362968c9fbe182a"}]},"branch":"refs/heads/master"},"3ad17e14430fa5ccd03f30c3580464d949573453":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-30 03:11:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/914777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/914777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/914777/2"}}},"commit":{"parents":[{"commit":"d94ecc22c036d060c0a52bd5130281e244a769e5","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d94ecc22c036d060c0a52bd5130281e244a769e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 03:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 03:10:45.000000000","tz":540},"subject":"Replace deprecated glanceclient exception","message":"Replace deprecated glanceclient exception\n\nExceptions without HTTP prefix were deprecated in glanceclient 0.4.0.\n\n[1] 354c98b087515dc4303a07d1ff0d9a9d7b4dd48b\n\nCloses-Bug: #2059821\nChange-Id: I9bce4bbeae78d8e4f95a5cd9de856edd54b05a76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ad17e14430fa5ccd03f30c3580464d949573453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ad17e14430fa5ccd03f30c3580464d949573453"}]},"branch":"refs/heads/master"},"713c2fbd568e4b31a80a6025fbd99d8f1f1dd07e":{"kind":"REWORK","_number":3,"created":"2024-03-30 11:05:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/914777/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/914777/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/914777/3"}}},"commit":{"parents":[{"commit":"d94ecc22c036d060c0a52bd5130281e244a769e5","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d94ecc22c036d060c0a52bd5130281e244a769e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 03:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 11:04:59.000000000","tz":540},"subject":"Replace deprecated glanceclient exception","message":"Replace deprecated glanceclient exception\n\nExceptions without HTTP prefix were deprecated in glanceclient 0.4.0.\n\n[1] 354c98b087515dc4303a07d1ff0d9a9d7b4dd48b\n\nCloses-Bug: #2059821\nChange-Id: I9bce4bbeae78d8e4f95a5cd9de856edd54b05a76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/713c2fbd568e4b31a80a6025fbd99d8f1f1dd07e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/713c2fbd568e4b31a80a6025fbd99d8f1f1dd07e"}]},"branch":"refs/heads/master"},"7322b01b6c3de5278e1beea1bd9e092d896a9cc4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-03-31 05:01:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/914777/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/914777/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/914777/4"}}},"commit":{"parents":[{"commit":"d94ecc22c036d060c0a52bd5130281e244a769e5","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d94ecc22c036d060c0a52bd5130281e244a769e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 03:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-31 05:01:57.000000000","tz":0},"subject":"Replace deprecated glanceclient exceptions","message":"Replace deprecated glanceclient exceptions\n\nExceptions without HTTP prefix were deprecated in glanceclient 0.4.0.\n\n[1] 354c98b087515dc4303a07d1ff0d9a9d7b4dd48b\n\nCloses-Bug: #2059821\nChange-Id: I9bce4bbeae78d8e4f95a5cd9de856edd54b05a76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7322b01b6c3de5278e1beea1bd9e092d896a9cc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7322b01b6c3de5278e1beea1bd9e092d896a9cc4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8a55382ce27110f754822072f106ff8d06bb47c3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-03-31 13:15:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/914777/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/914777/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/914777/5"}}},"commit":{"parents":[{"commit":"54856da91045299537fdb69edf43fb61aba79cc6","subject":"Merge \"PowerMax: Allow live migration without pool name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54856da91045299537fdb69edf43fb61aba79cc6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 03:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-31 13:15:46.000000000","tz":0},"subject":"Replace deprecated glanceclient exceptions","message":"Replace deprecated glanceclient exceptions\n\nExceptions without HTTP prefix were deprecated in glanceclient 0.4.0.\n\n[1] 354c98b087515dc4303a07d1ff0d9a9d7b4dd48b\n\nCloses-Bug: #2059821\nChange-Id: I9bce4bbeae78d8e4f95a5cd9de856edd54b05a76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a55382ce27110f754822072f106ff8d06bb47c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a55382ce27110f754822072f106ff8d06bb47c3"}]},"branch":"refs/heads/master","description":"Rebase"},"193ac168d74d867b84afda0bc949197c5953f6a9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2025-09-06 12:33:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/914777/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/914777/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/914777/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/914777/6"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-30 03:04:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-06 12:32:52.000000000","tz":540},"subject":"Replace deprecated glanceclient exceptions","message":"Replace deprecated glanceclient exceptions\n\nExceptions without HTTP prefix were deprecated in glanceclient 0.4.0.\n\n[1] 354c98b087515dc4303a07d1ff0d9a9d7b4dd48b\n\nCloses-Bug: #2059821\nChange-Id: I9bce4bbeae78d8e4f95a5cd9de856edd54b05a76\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/193ac168d74d867b84afda0bc949197c5953f6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/193ac168d74d867b84afda0bc949197c5953f6a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
