)]}'
{"id":"openstack%2Fcinder~915431","triplet_id":"openstack%2Fcinder~master~I2a69bda80dc142770d9551b3982808e28594fc25","project":"openstack/cinder","branch":"master","topic":"ruff","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-01 20:23:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a69bda80dc142770d9551b3982808e28594fc25","subject":"Update ruff config","status":"MERGED","created":"2024-04-10 14:47:43.000000000","updated":"2024-08-01 20:24:43.000000000","submitted":"2024-08-01 20:23:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915431-ruff","meta_rev_id":"5266b5b7443c516927e5a6c4ac553c301b06e94e","_number":915431,"virtual_id_number":915431,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-01 20:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2024-07-31 16:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2024-08-01 18:14:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-10 14:53:56.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-10 15:10:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-10 15:39:25.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-10 17:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-10 18:17:13.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-10 18:20:11.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-10 23:14:44.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-11 08:12:55.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-12 08:13:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 21:03:00.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-24 21:06:15.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-25 00:05:20.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-28 14:06:17.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-07-31 14:02:40.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2024-07-31 16:19:48.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-07-31 18:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"0c95d41c18cc9c9645382b860e62572b889d68a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-10 14:47:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db31923f0d522c8948ec91c4a24595a38a0b0a92","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-10 14:53:56.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/570d58d42f5d4852accfb931a09ecc59\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b7168b8fc1134fb5a03552df47cb6987 : NODE_FAILURE Node request 200-0000002959 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"4ad6dfe62665a140a9c0ed81e023780cd7235882","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 15:10:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea20dad3b9a415287da8ab8cce3fe88\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e27679c5ad14a3399f1b6fe48877968 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b4f6affeec94a3181d118c1d0a3c5c6 : SUCCESS in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a419196f1bcedc59f10dc96b6c19b73600b6f09f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-10 15:39:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/915431 : FAILURE in 50m 25s","accounts_in_message":[],"_revision_number":1},{"id":"770c959f3e43560c66df3646c006b5d25aa88685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 17:15:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea607f1f76904c7d85fccc6d6b8eac01\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fa135e5f887a48948ac7e3fea014502d : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0461fe651f474ab887eff26f97f28400 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba3ebb17c6834227937cd76f8bd60f8e : SUCCESS in 9m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58618632006d4e5b97f075c72f273d32 : FAILURE in 11m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e5d33d3a2df4689add3a37963c0c314 : FAILURE in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9708827669a4710a653d759475eb181 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/b2e330ea3f6247e0b19bd39404cce6aa : SUCCESS in 56m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59ed7cc3076d422f9094f1f3fbd06fc0 : SUCCESS in 1h 27m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da1d95001ad74f9191335d40b62a0262 : SUCCESS in 56m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50171ef7259a4911b73f13a2814bd4ac : SUCCESS in 16m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a89ff46877094d838b75e4cb1ba71d2e : SUCCESS in 14m 53s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bdb97c91c41d48d8a9b8009d788225c5 : SUCCESS in 13m 12s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47dfa7995e3544e8a927da74a5aa4173 : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe5194da7e14492ea6c96558298df2d1 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5dd33c9a1cb544a9b8a1029aa4b88763 : SUCCESS in 5m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a0c314dcb0644ad2a7913ed48f2d69d4 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/94a7c66d28f244ca8ac4029223051d14 : SUCCESS in 51m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed8aeda32dc94d0caf660a214a08ebfc : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a520352e05fd4817a4693d3660a9d674 : SUCCESS in 2h 13m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2e35b5a7cb1745ecbae6dfd9069898e0 : SUCCESS in 2h 03m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8c57251b8f774dcaba772763895d2955 : SUCCESS in 1h 56m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b25515ccc25242599c90ae7961837c79 : SUCCESS in 2h 21m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/825c616e12ae4037992889830b013a77 : SUCCESS in 25m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c76fefb2a6fb4253a0ee1187c0129fab : SUCCESS in 54m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b651a9bfba8c44798db764d762c96ecb : SUCCESS in 1h 10m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/258fb4c9baa04029ba751cef7b6263bd : SUCCESS in 37m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/390bbaf2609f4b9db1044338feea82b4 : SUCCESS in 1h 17m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5054a1345ba6464dbdc2beeb817be913 : SUCCESS in 1h 54m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ccaab56efd124d3ebc1be3fb6462c8d7 : SUCCESS in 1h 35m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce6106f5864142d8a216b472151e4309 : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b9f78eeea5e709916de72977e5d309a8fc746b05","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-10 18:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/195/.","accounts_in_message":[],"_revision_number":1},{"id":"44494a405f4ccd6bdead59e9574e30e2a8f367fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-10 18:20:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_Unity_FC/21e8898/DellEMC_Unity_FC/1647 : FAILURE in 1h 27m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_Unity_iSCSI/842fc38/DellEMC_Unity_iSCSI/1678 : SUCCESS in 1h 40m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"feaabb743745e4c4fd3164a3876c5173863e596b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-10 23:14:44.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-915431-1/FC/ : FAILURE in 8h 26m 39s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-915431-1/iSCSI/ : SUCCESS in 2h 7m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c3fcf75b162fdc2fe30b9c58aab62295382d9677","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-11 08:12:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerFlex_3.6/e760638/DellEMC_PowerFlex_3.6/1158 : SUCCESS in 1h 36m 24s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerFlex_4.5/c026172/DellEMC_PowerFlex_4.5/129 : SUCCESS in 2h 10m 37s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/1bec157/DellEMC_PowerFlex_4.x_NVMeTCP/34 : ABORTED in 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"029613df3292102b896497b2c121a8961a6a00c5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-12 08:13:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/915431/1/20240412085218/NEC-FC : SUCCESS in 4h 45m 08s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/915431/1/20240412085218/NEC-ISCSI : SUCCESS in 8h 21m 31s","accounts_in_message":[],"_revision_number":1},{"id":"10cec1acf7be18b24118ab490383b449a37769c1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-24 18:35:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f91be7b0cccb481f196fe3c797131825c70c97d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 18:55:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75ce6c1f0647455f9ce413e8c5ab3fca\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4fd297ed0f5a477c9549c9c8c63dcf62 : SUCCESS in 17m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d8b21e97c9ef4624ad8f7432428abd11 : SUCCESS in 17m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca691d31ac08b065a02924199d2bcc03542c9d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 20:56:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c951c0a47df4cc6996d91eb739e03d0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/926f064697b4487691dc4cac2c4e31e7 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926c50a9e17b45f0b699ab41ca328d73 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b564b9e0a9e4d8cb874e6d0d18921b7 : SUCCESS in 11m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4a7c64efd4c400f91d7406bdcde91d2 : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e98116d7823f49e88660517faa15d2b2 : FAILURE in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4ed5cd387c476f94cf95893b288fc3 : SUCCESS in 12m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/07455189381f4899ae293b71971b7b4c : SUCCESS in 49m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9740d832acc342b3ac473ef72b55e2cb : SUCCESS in 1h 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c164d3a589a44b0b259e0d0ae2d6833 : SUCCESS in 45m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33a62c9f280d4b51a779977d9fb92d72 : SUCCESS in 10m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/963f1d9628da4a59a152528ae9b54ed1 : SUCCESS in 16m 47s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d25c4f4175d94ba0877b3505272c33c3 : SUCCESS in 14m 03s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f68d2fd084304ce3b93d070aae15d386 : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b249725e01fd45d18f40fa1dd8e84e62 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22abaa37f8e04885809d727f4dac52f6 : SUCCESS in 6m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/39d5a48ff6274f1fbcea373fffe4fd6c : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/55d20661581b4d05ae573bdbe303682e : SUCCESS in 45m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad87bb3260f244a4b7af7c139ecc8118 : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1455be29369c4caa863e10c780b99589 : SUCCESS in 2h 06m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd04a87d4c85416e86ae947304e8fa67 : SUCCESS in 2h 13m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/07315b1d884f4eb4955c67d45a1c999c : SUCCESS in 1h 42m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c48a922564e9413ca9fb9a50adf50f40 : SUCCESS in 2h 01m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cd06713188e34c42a29ed2430b4b5137 : SUCCESS in 38m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e440b0788264ca5bc8c8af3e4195ed4 : SUCCESS in 1h 15m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4a6f8261a5524c90a8938015b8ec65db : SUCCESS in 1h 07m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8069e1f4ce704ff38e8fee7053c8a48c : SUCCESS in 25m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3d29af51ce5742129159c407eed4b48f : SUCCESS in 1h 01m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/009fb32337b94d8aa834507fca9cfd00 : SUCCESS in 1h 05m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb76319c966048c3a420930bf051e46c : SUCCESS in 1h 11m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/917249b5ac6643c38d4b99d98d22ed31 : SUCCESS in 49m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3b6b5a9274c2250ece0f31da997cf89dbc4aa703","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-24 21:03:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_04_2024_20_56_22\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"a543bbaf645a8a0c6435c32be3fd2e37efce713e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-24 21:06:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/915431/1/upstream-check/cinder-tempest-plugin-ontap-fcp/03d0d6d/ : RETRY_LIMIT in 3m 53s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/915431/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/893d6d8/ : RETRY_LIMIT in 4m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/915431/1/upstream-check/cinder-tempest-plugin-ontap-nfs/dfe60f1/ : RETRY_LIMIT in 3m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/915431/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f212888/ : RETRY_LIMIT in 3m 45s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"cf493ebad4c721d8e00803bd5fb58b7c6baa08f8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-24 22:27:09.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/636265f13e9047fe9bc05f10dbf0bf57\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f17c55bb43944624912ff42baea7408e : FAILURE in 1h 25m 15s","accounts_in_message":[],"_revision_number":1},{"id":"40ed545e058fb7413ef14c6b0145c36591223c17","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-25 00:05:20.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerStore_iSCSI/a810091/DellEMC_PowerStore_iSCSI/1271 : SUCCESS in 1h 11m 52s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerStore_FC/34d35c6/DellEMC_PowerStore_FC/1442 : SUCCESS in 59m 20s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerStore_NFS/dd02bc7/DellEMC_PowerStore_NFS/961 : SUCCESS in 1h 08m 54s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerStore_NVMeTCP/d6b7460/DellEMC_PowerStore_NVMeTCP/822 : SUCCESS in 1h 38m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"9b6efa976826c5797ae3f46d631fb641e4766f6e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-25 14:19:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_Unity_FC/33fbdd1/DellEMC_Unity_FC/1766 : FAILURE in 1h 33m 49s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_Unity_iSCSI/921b1a4/DellEMC_Unity_iSCSI/1795 : FAILURE in 1h 50m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1644cb5b9dcb3eca3bfe7689928633050e1eaf57","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-10 09:54:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerFlex_3.6/d267e2b/DellEMC_PowerFlex_3.6/1266 : SUCCESS in 1h 41m 24s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerFlex_4.5/442a8af/DellEMC_PowerFlex_4.5/201 : SUCCESS in 2h 01m 14s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/915431/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/1d01c2a/DellEMC_PowerFlex_4.x_NVMeTCP/141 : FAILURE in 2h 53m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"acadd0ec255808d941e08579604b284e760738d9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-28 14:06:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/915431/1/20240728190508/NEC-V-FC : SUCCESS in 3h 52m 23s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/915431/1/20240728190508/NEC-V-ISCSI : SUCCESS in 4h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9c7a901a2c53de42f02f297386646f08bb108df1","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-07-31 14:02:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"719763f940fbfe92ccc49135cf15d36fd6d04cad","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-07-31 16:19:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b046a3469488c2af6ef17cf0558cb15f3419e20","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-31 18:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/915431/1/fc : SUCCESS in 0h 47m 47s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/915431/1/iscsi : SUCCESS in 0h 57m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b1d1cbfa8e673df186add67aa3a409267a63a481","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-08-01 18:14:22.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e97f28df3bf8c1d9e5bf07540e1da33a2aa1062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 18:14:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d71eb22439f7c78de71955855105f86eae9a172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 20:23:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/844752b5a1f74c63b138d2ecaa8b207d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/92b5d4f3a45440199cecce3290371341 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d81f8b1e6e4a85a9dc8ebbfd00ed56 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adee3e1279f8429295fe752ff807ae31 : SUCCESS in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/088c412f632a4c27b3e51e81a982f61d : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25c2dc03dda4d17a4c8b6b07accebfd : SUCCESS in 17m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/67f62915755c4b4dabf561ede96c1738 : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9987d41c9034694a0e04f2952c515a2 : SUCCESS in 1h 28m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f13cd4cd8a954db1989d2e08a84c5c9b : SUCCESS in 58m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e39c052a2cd44a1b33496a410645caf : SUCCESS in 20m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e080badeb1274419b008e03c5539d51c : SUCCESS in 1h 14m 59s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f202b2b3fbaa422ab42ae63481c9ef91 : SUCCESS in 2h 08m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e081ed8dba3485c84be8e82c32aa92f : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4b629de3e505978239a164b5855226f9f5922762","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 20:23:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5266b5b7443c516927e5a6c4ac553c301b06e94e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 20:24:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c325aadc4a264658bd556b00a298b3e2\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/f7c9148989ab499699481309cec88297 : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/934b8e12d96e430e8671a9af6a800580 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2532e4c9f9514fe7850f43af2e7ced9c : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f51bbbb30269afd91db90ea4764c18b7f61b49a","revisions":{"8f51bbbb30269afd91db90ea4764c18b7f61b49a":{"kind":"REWORK","_number":1,"created":"2024-04-10 14:47:43.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/31/915431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/915431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/915431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/915431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/915431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/915431/1"}}},"commit":{"parents":[{"commit":"b0f0b9015b9dfa228dff98eeee5116d8eca1c3cc","subject":"Merge \"hacking: Fix E501 errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b0f0b9015b9dfa228dff98eeee5116d8eca1c3cc"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-04-10 14:41:43.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-04-10 14:42:25.000000000","tz":-240},"subject":"Update ruff config","message":"Update ruff config\n\nruff 0.3.5 issues warnings for these, update to current form.\n\nChange-Id: I2a69bda80dc142770d9551b3982808e28594fc25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f51bbbb30269afd91db90ea4764c18b7f61b49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f51bbbb30269afd91db90ea4764c18b7f61b49a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
