)]}'
{"id":"openstack%2Fcinder~916044","triplet_id":"openstack%2Fcinder~stable%2Fzed~Id239b4150b1c8e9f4bf32f2ef867fdffbe84f96d","project":"openstack/cinder","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-04-29 11:20:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id239b4150b1c8e9f4bf32f2ef867fdffbe84f96d","subject":"cinder-backup: use the same backup backend host for incremental backups","status":"ABANDONED","created":"2024-04-17 07:40:46.000000000","updated":"2024-04-29 11:20:12.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eb4077479f40cb547cc39bffeb4366b6133ec23d","_number":916044,"virtual_id_number":916044,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-17 20:37:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 07:40:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"state":"CC"},{"updated":"2024-04-17 07:50:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-17 07:55:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-17 09:06:18.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-17 09:58:23.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-17 10:30:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 13:32:57.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-19 14:03:18.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-04-19 23:13:36.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-29 11:20:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"2a3c678feead5107e4e477d4774fbb27e258d41a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-17 07:40:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08d87e0db0fb5d7971aa8d22ce6c3bcc5471a6f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 07:50:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cce431e64f24c29b90279658c00c7d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/039f83e1a9fc45a0a38b48fb48567bfa : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/632f90b586aa4d0d8ed58117b6eac26d : SUCCESS in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eea592e334e480262a42caabf0f7839600b7d0e6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-17 07:55:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916044 : FAILURE in 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d4cfbd0ba650509a34a534b10cccc556d82662a6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-17 09:06:18.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2c193b98c3144f11871c3b78e1a28b55\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/162eaad9873146b4b2b7c9b993abb928 : FAILURE in 1h 21m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f586e90b5f77625a6c91ffa08b0cd7366bb8c575","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-17 09:58:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/44/916044/1/ : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":1},{"id":"36f6d0df56404cf4b901368eb099ddd02977d8f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 10:30:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40b6003880624b0c95e6904d4862ed81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500288e4d11a4415b12d7cb9a87a3027 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/882e54ab06a644bfbcdd1e6577e8c9d5 : SUCCESS in 13m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8100a6be8ac34c42b4d4c3a7967b5f51 : SUCCESS in 10m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/233bd7e5f0624ca6b1bdcd6b8f149b65 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657df28a5e4f4694b3e9727b90e474f4 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61917873400a40829c60752f7c8c31e9 : SUCCESS in 15m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/435c08a8db9d4aa889cee6fc7aa04928 : SUCCESS in 16m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ecf9ab07427840388d303aabca4ef7c1 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc8ec93eae694e05be2e1238bf818771 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b5ab496ad42e46efa9abef524618c028 : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e3e4b5c19f2440185d71b54cb0a2c99 : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/046ecf335c3a46cd8dd2d6763661deff : SUCCESS in 40m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14a9fa19c8fa4b57ab1cb7fcd71174f6 : FAILURE in 7m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a93d7ed4ff294135b05855c326b3902b : TIMED_OUT in 2h 04m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8a8942ac33504ed3a5d7cf9e1f92aaaa : FAILURE in 1h 28m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/627e470b26ac467986fe246fc9b2a3e1 : SUCCESS in 1h 15m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e21e19510b04427b23ff4ca08f0fd06 : SUCCESS in 1h 10m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2fbc67832e654f0b9116d38c485e00a6 : SUCCESS in 46m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d5f5a8df644b4f9bb41f6e263f9e6bce : SUCCESS in 1h 12m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dec94a8ee8f1408bbbcabccfcaed5059 : SUCCESS in 1h 15m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe646c88b7b84f3da901856661a8d71a : SUCCESS in 50m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/562c07a02e1d450286e985a8871651eb : SUCCESS in 50m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2163f8b312194300abc1cb6eac9680ea : SUCCESS in 43m 05s","accounts_in_message":[],"_revision_number":1},{"id":"05d4323c50d55c2582f330f87aaff93e7962d5f8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-17 13:32:57.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-44-916044-1/FC/ : FAILURE in 5h 51m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-44-916044-1/iSCSI/ : SUCCESS in 1h 26m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"078984ed23dd8447f2428c9abdf7df754875240f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-17 18:46:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8881bc257e7811d72ceb1c9ab327f91a3960545","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 18:58:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9608951eeab40e49b0220fba3deedb9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0b924ec1eab4eebb662beffbdd58278 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/183232d96799480b8d0f613dd58d78d8 : SUCCESS in 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3d2a6ffe6014128786a654a53ec850c3a4fe71a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-17 20:08:28.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8017f72f7b4340cfb773f09041fbd9ce\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/cb8b1ebdc3134796a020fe66f813ac42 : FAILURE in 1h 18m 44s","accounts_in_message":[],"_revision_number":1},{"id":"eb4b9ab4e3bda6546a1778359c88cec9ce4f1703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 20:37:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9492f22ecc054eb283fc8506a79cde87\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60014afc5c244b88b6c1da2f41ab4883 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/603f7d10f62b4a349308b534e2e00bcc : SUCCESS in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e85b1cca6d4449ff93b9fec968dea378 : SUCCESS in 10m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1213fb93149423bad131a360077c684 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beaed884a4614695a1bdc9067e93dada : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/337a1d8e08284853a282c73e3d4cff1a : SUCCESS in 15m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28a5068142494ed98dc130079bb3f446 : SUCCESS in 15m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/adc356604b994132ae46ef88452c51c4 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/37b475b7190e41249d384b5f314ac872 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/601a689f66be450f964d20dc1a143575 : SUCCESS in 5m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85209b12e4a640b4ac24198c44989e21 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2bba96f7efc445aea96ac6200c0f38c6 : SUCCESS in 46m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d6d47eb72244bb092aad335dc01a76f : FAILURE in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6228ef2401d741f69ada5c6cfe908843 : SUCCESS in 1h 43m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01aa3330b4254eeba976285d0c207975 : FAILURE in 1h 19m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89d9fa3a85ab4c71adf5b26bdeb27c2a : SUCCESS in 1h 06m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f1fb3eeb58f442997284f943a2be403 : SUCCESS in 1h 15m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5f68327d2e7405fa5fd82f8190c26d5 : SUCCESS in 41m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a8766f7a4c704c108cf97f9a7f93b5db : SUCCESS in 1h 06m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b90d1a09378a4c49b3372f276314438a : SUCCESS in 1h 22m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb696a7f95aa478486bfc0e81c3f973f : SUCCESS in 1h 04m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/949c266320d64ed5bca78518f88a9427 : SUCCESS in 37m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1d750447ee64d89964b789d9d694ad9 : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b0da587cc18d38c74b02fccfc267c0451455c910","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-04-19 14:03:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/44/916044/1/20240419193536/NEC-V-ISCSI : SUCCESS in 2h 59m 23s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/44/916044/1/20240419193536/NEC-V-FC : SUCCESS in 3h 27m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ca100bcab58b0ff6eae6dfb5d02bbdd6d911d5b8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-19 23:13:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/44/916044/1/20240420000502/NEC-ISCSI : SUCCESS in 3h 58m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/44/916044/1/20240420000502/NEC-FC : SUCCESS in 8h 08m 30s","accounts_in_message":[],"_revision_number":1},{"id":"eb4077479f40cb547cc39bffeb4366b6133ec23d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-29 11:20:12.000000000","message":"Abandoned\n\nstable/zed branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/zed if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"396ec95b060ffcc2cb1a27e682e2ab0d71ac86a3","revisions":{"396ec95b060ffcc2cb1a27e682e2ab0d71ac86a3":{"kind":"REWORK","_number":1,"created":"2024-04-17 07:40:46.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/916044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/916044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/916044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/916044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/916044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/916044/1"}}},"commit":{"parents":[{"commit":"f74e2729554bee01b0a3e631a8001bb39e540433","subject":"Ceph minimum client on cinder-plugin-ceph-tempest job enable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f74e2729554bee01b0a3e631a8001bb39e540433"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2022-10-17 03:37:36.000000000","tz":-300},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-04-17 07:40:20.000000000","tz":120},"subject":"cinder-backup: use the same backup backend host for incremental backups","message":"cinder-backup: use the same backup backend host for incremental backups\n\nIncremental backups only work if there\u0027s a previous backup to base\nit on. With the posix driver, this means there needs to be a previous\nbackup on the same host where the incremental backup is created.\n\nThis patch ensures that an incremental backup is scheduled on the\nhost that that contains the base backup being used for the increment.\nPreviously we were relying on luck for this and much of the time an\nincremental backup would fail due for want of a base backup.\n\nCloses-bug: 1952805\nChange-Id: Id239b4150b1c8e9f4bf32f2ef867fdffbe84f96d\n(cherry picked from commit 73c0d73aba1855edb774d69ae940a1b53dc975e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/396ec95b060ffcc2cb1a27e682e2ab0d71ac86a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/396ec95b060ffcc2cb1a27e682e2ab0d71ac86a3"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
