)]}'
{"id":"openstack%2Fcinder~916056","triplet_id":"openstack%2Fcinder~master~Iafa5c8240101f357c2776814d7a4e0f600c21cbf","project":"openstack/cinder","branch":"master","topic":"remove-stale-deps","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-04-17 14:27:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-17 14:27:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iafa5c8240101f357c2776814d7a4e0f600c21cbf","subject":"backup: Remove use of oauth2client","status":"ABANDONED","created":"2024-04-17 10:54:44.000000000","updated":"2024-04-17 21:13:23.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"15b1dcfc8eba3e99da7c02b17969219cadf4db4a","_number":916056,"virtual_id_number":916056,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-17 13:23:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 11:06:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-17 11:28:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-04-17 13:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 13:32:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-17 14:22:12.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-17 17:22:52.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-17 21:13:23.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"}],"messages":[{"id":"ecb86ba825e9ea8e5af632575a69ac7a33145691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 10:54:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f9d3718a6a059a28fc1e9500852b5cf1bc6f4d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 11:06:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419ec26b23db405c825a3c20218eced5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59eddadf4f43439fa065f0867cdfe2ec : FAILURE in 5m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/83302f220f3744dc99206c589c54a50a : FAILURE in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2656707b4a4467e4d33fec3bedfcc24029f5cb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-17 11:28:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"152c22e19d5cbf792b915b32a1a502bf2ecbe117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 13:23:28.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dba6d25b858c4e7284c60ddc9cba6f61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb9e1cca05948f2b7f71af3cb090d0e : FAILURE in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebcfdc48a53b427ca310a8b7146f2cea : FAILURE in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66520dd54b0e4c258de6603a44368fea : FAILURE in 6m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ee731280efe4e798045e0424c62ca0e : FAILURE in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e32818436354394945e7244717e5596 : SUCCESS in 12m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d722a098cdd4456a4cfb4d530d577ad : SUCCESS in 2m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/88e65e176e384e33af5c94d6fb87461d : SUCCESS in 57m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69bb37b7f72b4a0db81c919589860437 : SUCCESS in 1h 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01bbd93dce4d4c158faa1ccb61ebcdb4 : SUCCESS in 52m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/595d006a78794c179b0537c24577555b : SUCCESS in 15m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e2646c81f344eaea02614fd55dbe3b0 : FAILURE in 4m 38s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/39d3a34722224a43945dc9f74b2a4856 : FAILURE in 7m 55s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/19599e7263c6485e9f8540e59e2f2463 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a8639a9779c4d29b250e01ec0b39535 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/386356b95362428987fb1f0bd47b175a : SUCCESS in 5m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d63d37a4773a463ba2d9d87a8b17125b : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13d5b4c62f464f4f8185a2ff5932b48e : SUCCESS in 37m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/74b1f51a2ff34ddcb0f2566b4762e3fd : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ebf129baf6e543dcae41759e5ae2cd11 : SUCCESS in 2h 00m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4aacdfec6adf45ff85cf280ef1807ca1 : FAILURE in 2h 07m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0cdf580c08ae458ba5fd14d6188ada69 : FAILURE in 1h 45m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9602ee85f30646f7ab944281450f049d : SUCCESS in 1h 43m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/daa8727867eb4bbcb7bd86046a85a7b3 : SUCCESS in 31m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a52695abb80d4ee4814048b78905cc15 : SUCCESS in 1h 08m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5add99d10fb432daa56db9aaa2b6cc2 : SUCCESS in 1h 41m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76d063dc8c784e9ead4fad9c91606fe5 : SUCCESS in 51m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2206cab9f0a046e483dbe7086956602f : SUCCESS in 1h 23m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c55d6d81fc8d470a8f0ab1e9e1332626 : SUCCESS in 1h 20m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c692f6cfedd24e7ba7157e2b089f91ed : SUCCESS in 1h 41m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ecedddd932940f88c42006f59ca0fd7 : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cb544ed213d88548bd0c2a8186db50c08307c3de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 13:24:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dba6d25b858c4e7284c60ddc9cba6f61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb9e1cca05948f2b7f71af3cb090d0e : FAILURE in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebcfdc48a53b427ca310a8b7146f2cea : FAILURE in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66520dd54b0e4c258de6603a44368fea : FAILURE in 6m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ee731280efe4e798045e0424c62ca0e : FAILURE in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e32818436354394945e7244717e5596 : SUCCESS in 12m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d722a098cdd4456a4cfb4d530d577ad : SUCCESS in 2m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/88e65e176e384e33af5c94d6fb87461d : SUCCESS in 57m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69bb37b7f72b4a0db81c919589860437 : SUCCESS in 1h 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01bbd93dce4d4c158faa1ccb61ebcdb4 : SUCCESS in 52m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/595d006a78794c179b0537c24577555b : SUCCESS in 15m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e2646c81f344eaea02614fd55dbe3b0 : FAILURE in 4m 38s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/39d3a34722224a43945dc9f74b2a4856 : FAILURE in 7m 55s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/19599e7263c6485e9f8540e59e2f2463 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a8639a9779c4d29b250e01ec0b39535 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/386356b95362428987fb1f0bd47b175a : SUCCESS in 5m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d63d37a4773a463ba2d9d87a8b17125b : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13d5b4c62f464f4f8185a2ff5932b48e : SUCCESS in 37m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/74b1f51a2ff34ddcb0f2566b4762e3fd : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ebf129baf6e543dcae41759e5ae2cd11 : SUCCESS in 2h 00m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4aacdfec6adf45ff85cf280ef1807ca1 : FAILURE in 2h 07m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0cdf580c08ae458ba5fd14d6188ada69 : FAILURE in 1h 45m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9602ee85f30646f7ab944281450f049d : SUCCESS in 1h 43m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/daa8727867eb4bbcb7bd86046a85a7b3 : SUCCESS in 31m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a52695abb80d4ee4814048b78905cc15 : SUCCESS in 1h 08m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5add99d10fb432daa56db9aaa2b6cc2 : SUCCESS in 1h 41m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76d063dc8c784e9ead4fad9c91606fe5 : SUCCESS in 51m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2206cab9f0a046e483dbe7086956602f : SUCCESS in 1h 23m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c55d6d81fc8d470a8f0ab1e9e1332626 : SUCCESS in 1h 20m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c692f6cfedd24e7ba7157e2b089f91ed : SUCCESS in 1h 41m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ecedddd932940f88c42006f59ca0fd7 : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":1},{"id":"aafa99e2beb006a954074427b33db3af0cdedd0f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-17 13:32:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916056 : FAILURE in 44m 38s","accounts_in_message":[],"_revision_number":1},{"id":"91ab7550d74e4453eb4a8e48426c4e6fb25d4164","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-17 14:22:12.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/49a74ca62a8642c8bed95d6eac9c870f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6a6dffbefa0a4cb1b8032345383eb655 : FAILURE in 1h 24m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e46637a7033e956fe8a4f4bd0dfb51b3b254520b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 14:27:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60661efac195148a3552259c1935ab0aca8e792c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 14:27:46.000000000","message":"Abandoned\n\nDupe of existing change https://review.opendev.org/c/openstack/cinder/+/902120","accounts_in_message":[],"_revision_number":1},{"id":"1e741370eb3a4c79e5dd6e1c46d92d203bd146ab","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-17 17:22:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/232/.","accounts_in_message":[],"_revision_number":1},{"id":"15b1dcfc8eba3e99da7c02b17969219cadf4db4a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-17 21:13:23.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-56-916056-1/FC/ : FAILURE in 8h 19m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-56-916056-1/iSCSI/ : SUCCESS in 3h 12m 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d558b85326015ac248e5e08d7b6e28710982a67","revisions":{"0d558b85326015ac248e5e08d7b6e28710982a67":{"kind":"REWORK","_number":1,"created":"2024-04-17 10:54:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/916056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/916056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/916056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/916056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/916056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/916056/1"}}},"commit":{"parents":[{"commit":"b36142602e4a478e0e9349126ab2bee92f3f2d18","subject":"Merge \"Fix snapshot status is always backing-up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b36142602e4a478e0e9349126ab2bee92f3f2d18"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-17 10:45:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-17 10:53:05.000000000","tz":60},"subject":"backup: Remove use of oauth2client","message":"backup: Remove use of oauth2client\n\nThis is deprecated and hasn\u0027t seen a release in nearly 6 years [1].\nMeanwhile the version of google-api-python-client that contains the\nrequired functionality, 1.6.0, was released over 7 years ago [1]. We no\nlonger need the fallback.\n\n[1] https://pypi.org/project/oauth2client/#history\n[2] https://pypi.org/project/google-api-python-client/1.6.0/\n\nChange-Id: Iafa5c8240101f357c2776814d7a4e0f600c21cbf\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d558b85326015ac248e5e08d7b6e28710982a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d558b85326015ac248e5e08d7b6e28710982a67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
