)]}'
{"id":"openstack%2Fcinder~916434","triplet_id":"openstack%2Fcinder~master~Ia3104b702839688852fc535f01b29018bdf3c9a5","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-30 20:16:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3104b702839688852fc535f01b29018bdf3c9a5","subject":"api: Move rate limiting middleware to its own module","status":"MERGED","created":"2024-04-19 14:51:42.000000000","updated":"2025-10-30 20:16:38.000000000","submitted":"2025-10-30 20:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916434-openapi","meta_rev_id":"f561a880dcd689a39c646ce8b45bd350a01377f7","_number":916434,"virtual_id_number":916434,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-30 20:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-24 12:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-10-30 14:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-10-30 14:38:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 15:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 17:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 19:14:57.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 19:57:03.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-19 22:00:11.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-04-19 23:16:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-20 02:02:41.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-22 13:42:32.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-23 15:13:03.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-23 15:43:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-23 16:24:20.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 04:27:44.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-04-25 02:32:47.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-26 00:40:57.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-04-26 02:18:21.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-08 22:23:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-20 11:45:48.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-09-18 14:15:08.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 16:51:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-18 18:28:52.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-18 23:32:44.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-25 00:56:57.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-02-21 17:02:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-24 13:46:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-30 14:38:26.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-10-30 14:38:42.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"1eb0238a1e3771570f78c5b2c6d4aba61597ed45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 14:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec5732272ca2c554090d06e38ea024b2352c963d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:05:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7886bca207f148b1b19ffcf48cf103ae\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fcb0377efb149e0be31a724f0e560b0 : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28ec6757d5504da7a88b3508513db76b : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9fcfab35579f8bc22531b966222d9f5847a359c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:23:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0face69ea4d440db1acab98f357abe1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712f7f85efb448839456faf4786fbde4 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cde4748e0cec481c97d635b7d7c3ec9a : SUCCESS in 12m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/186c52b230b5441589dcdad2a4bcb48d : SUCCESS in 9m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9649f080ef3d47a0a5064c6d3cb908bf : FAILURE in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/746cc7e327ba4cac8a1f7c0a1dc01892 : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a7819abb732840e9a404ff7c8682b4ba : SUCCESS in 55m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fcfc717848ef443985e91ca8897bb171 : SUCCESS in 1h 33m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0170f84c76e24a138a68839bf7746366 : SUCCESS in 1h 04m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4bd3e35990cb4e74b8962c367db5d07e : SUCCESS in 18m 54s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6675f627866f492184b76e2d25ba4cdb : FAILURE in 14m 11s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5b92bfd2d9c44f5688ec732a23655077 : SUCCESS in 6m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/342829a02bb24ab3ba1271ca961ce6fc : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5584804dc2ad4f33a6fa12b212f64055 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8ea7d3d4f6f4d7598f90c794d1d129b : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4dca0de1855f4d0d984b784c17f770c2 : SUCCESS in 45m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa1b82820a09418badd9d8a0fd8d246d : SUCCESS in 7m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cb12aba97d954b3cbc993d08a9d5d4e9 : SUCCESS in 1h 49m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/192815f02aae4befb2b9790913ca2324 : FAILURE in 2h 16m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a99359f13ee2491aa5b4870ea630be14 : FAILURE in 2h 21m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9b6e67c05e434176a51c8d4ef8cd8970 : SUCCESS in 2h 02m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/be1506baed42431ba0c80d0cdc9ae920 : SUCCESS in 31m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2acadcd2cbdf475295e808a2f7938348 : SUCCESS in 1h 04m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3dd1872370554e57924d30da4d75397f : FAILURE in 1h 36m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f0469bad8a0b4688acc2c5feaa595b88 : SUCCESS in 34m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/38195bcf51fa4a21b61319ca97b05141 : SUCCESS in 1h 36m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/001ef3bf52c94bd1a2cd1cfeac84b5d1 : SUCCESS in 1h 38m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c81ee200472348f994049868d5abd75c : SUCCESS in 1h 26m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/925bda8e3a144559b784e7ad0cbdd7b1 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1e0e799f1a375fdbb50d0e80e67aac9d205d7c88","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-19 19:14:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916434 : FAILURE in 36m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b7e0695e76adbd9ee8a6bfb0513bcc4bced8b6eb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-19 19:57:03.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9e0712e7c0d84713b38ba635cf5881c5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/26a7ac641ae744f29479d49d3903c921 : FAILURE in 1h 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4f1eb545bbd7bb22eb73a46f398ba3e4390b424b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-19 22:00:11.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/916434/1/fc : SUCCESS in 0h 58m 03s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/916434/1/iscsi : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"919788bc10b35a0d157420f57c5121dd2791b955","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-19 23:16:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/916434/1/check/DellEMC_Unity_FC/25c1e39/DellEMC_Unity_FC/1706 : FAILURE in 1h 21m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/916434/1/check/DellEMC_Unity_iSCSI/7e2712f/DellEMC_Unity_iSCSI/1736 : SUCCESS in 1h 30m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"71a30a5718b49daca1a97db05fcef42e5c578fea","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-20 02:02:41.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-916434-1/FC/ : FAILURE in 9h 19m 9s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-916434-1/iSCSI/ : SUCCESS in 4h 32m 9s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"b1249d3139416a64568e01c8c941b6091b4058d8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-22 13:42:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/247/.","accounts_in_message":[],"_revision_number":1},{"id":"5e32515daa0c21a014320a1bab8f436cb7f96686","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 11:06:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c65d9a3366afcf9bba1bf86e0e32a4ff7bad42c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 11:28:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a81cb70271bb4506867dcd4954f2086e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56bf8a37957c4ab8bc13af7f84d3a848 : SUCCESS in 20m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12916c838dd94e558818277030243d01 : SUCCESS in 14m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8d7bb0a2ee82401137492c39d51c14994b95091","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 11:55:29.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/624b6d5ff63e4cb5a6213a42b304ea27\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e1273200bab5450abdae9908382d875e : FAILURE in 40m 36s","accounts_in_message":[],"_revision_number":1},{"id":"df991db256189df84118fcad4c7b4ae12dd55909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a98dd798c22b82a8b26975973e16fe864ec97f0b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 13:08:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916434 : FAILURE in 13m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2156781192f297724ffc8f87c6ab744e1abf2705","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:15:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a489d7bf03ea48328d89786306ecdea3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcb29af0d4964193a6293cfcea5738e0 : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6d4f0752f7ab467d8b27cecd7e903ca0 : SUCCESS in 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa325c29325e0f97552c926e1ea0f46232e9dd85","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 14:32:38.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2fade871fc8e46bb886b8ed20928f89a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/952b9defb9b843c7915b000c0aba605d : FAILURE in 31m 07s","accounts_in_message":[],"_revision_number":2},{"id":"92da955fd0ae5da1ba95ae50f733da7ebfb477ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:06:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76a3dceb1810401cb83beb936d006a24\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4807b593a245d996fdb1914470d560 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/449b2fd973244695a6f405791b886057 : SUCCESS in 10m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/398dd7bc78874d03a56623b174a8f921 : SUCCESS in 11m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fed7a4b7146f4959835a4b8cefbecd54 : FAILURE in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb97d6e1c5f849688335b2fa323b176f : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/13f3b2d332fe4a40afc345e769302155 : SUCCESS in 59m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3af9ff420996472ca915b0664f91841c : SUCCESS in 1h 27m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddf2fcc8a0354571bae58a413d35f0dc : SUCCESS in 50m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/686e1579c1084501ad81d10072748e4f : SUCCESS in 15m 41s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b58c0b80cfca499c814bba290f6d1f14 : SUCCESS in 13m 19s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fd4b8e088f0f41babed92d2211f10dd6 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfccb84688fc40859417d608ee549496 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf42b8635bed4371a128125b8dd6d460 : SUCCESS in 6m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6151ea2aa3e2432f9394e7e7495c8549 : SUCCESS in 7m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6dcb041f3fa347b8a353fa5189b5c98a : SUCCESS in 49m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/526878fdccba4934a7db5775e383ad0b : SUCCESS in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8e2537bc6a64398b2554507a924c22d : SUCCESS in 2h 03m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e7d175123534f569c224b7bc03578d6 : FAILURE in 1h 52m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db968fc0720245e38a42c208df6dad85 : SUCCESS in 2h 07m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e6dd17cba8e44ef28d440cfa1e85b70c : SUCCESS in 1h 55m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2f1a834b016e48d1a0015dcfeb8409eb : SUCCESS in 41m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bf1834dd004b4c9f9c63d681272ba7f1 : SUCCESS in 1h 16m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f2af6797af644ad990a267e77acabc77 : SUCCESS in 1h 38m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1439c390a53b4b208abe02205630f20f : SUCCESS in 40m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d36cba8bddc541e6acc51b3928fb20c4 : SUCCESS in 1h 28m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f20a3bedb2f840bca8593ddaedc4dbbf : SUCCESS in 1h 12m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc57eb7998554c7cb40f3b285be72c92 : SUCCESS in 1h 43m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8652436b5d84ef69a3f591d4149122a : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":2},{"id":"f4b61d28251372bfd04db48dad34daf51e71f48a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-23 15:13:03.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_04_2024_15_06_51\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"27d154a244642bfbd8e17d14633b166e351be83f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-23 15:43:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/2/upstream-check/cinder-tempest-plugin-ontap-fcp/11dd6ef/ : RETRY_LIMIT in 4m 23s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/45e846b/ : RETRY_LIMIT in 4m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/2/upstream-check/cinder-tempest-plugin-ontap-nfs/0602bbb/ : RETRY_LIMIT in 3m 40s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/6ee7d4a/ : RETRY_LIMIT in 4m 03s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cb66d40b1f936b2a1dc187924a800038f264cd28","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-23 16:24:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/34/916434/2/check/fujitsu-eternus-dx-iscsi/51ce3f9 : FAILURE in 3h 30m 12s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/34/916434/2/check/fujitsu-eternus-dx-fc/2a1a0c4 : SUCCESS in 3h 29m 32s","accounts_in_message":[],"_revision_number":2},{"id":"96aa901c7cf569bf2e2af3c0d7374593c4f6e604","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-23 19:23:52.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_Unity_FC/2e82151/DellEMC_Unity_FC/1741 : SUCCESS in 1h 20m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_Unity_iSCSI/2b89260/DellEMC_Unity_iSCSI/1771 : SUCCESS in 1h 34m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"cc618f06d42933d2b9f94109a0a867aff435b2b2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-24 00:34:08.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-916434-2/FC/ : FAILURE in 10h 10m 27s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-916434-2/iSCSI/ : FAILURE in 6h 8m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"ee7c13787639867be12540ccbca7bdd0e7567a03","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-24 04:27:44.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_PowerStore_iSCSI/84fc071/DellEMC_PowerStore_iSCSI/1265 : SUCCESS in 1h 06m 22s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_PowerStore_FC/44df8e7/DellEMC_PowerStore_FC/1436 : SUCCESS in 56m 11s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_PowerStore_NFS/ea2c409/DellEMC_PowerStore_NFS/955 : SUCCESS in 1h 09m 20s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_PowerStore_NVMeTCP/534bc6c/DellEMC_PowerStore_NVMeTCP/816 : SUCCESS in 1h 39m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"0bc3d877ec9040d4af170741486b95f75572433f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-24 12:24:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/281/.","accounts_in_message":[],"_revision_number":2},{"id":"37dd4b319b910ee6714e031075850bdfa9d1896d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-25 02:32:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/34/916434/2/ : FAILURE in 1h 38m 58s","accounts_in_message":[],"_revision_number":2},{"id":"df47ecf22d80d2e7bca5438c3557684e3dc6fd08","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-04-26 00:40:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/2/20240425222712/NEC-V-ISCSI : SUCCESS in 3h 07m 50s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/2/20240425222712/NEC-V-FC : SUCCESS in 11h 13m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1c1682a20c59ef6a28a1befefabe804b877a587c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-26 02:18:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/2/20240426063206/NEC-ISCSI : SUCCESS in 4h 19m 17s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/2/20240426063206/NEC-FC : SUCCESS in 4h 46m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0f1aa0c183934ff864804cd2de5352fb4506a6ea","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-27 00:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/916434/2/fc : SUCCESS in 0h 52m 38s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/916434/2/iscsi : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":2},{"id":"bbbeba49b680ecea7535e398591914cb3ae36cb4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-08 22:23:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_PowerFlex_3.6/74b2f7f/DellEMC_PowerFlex_3.6/1243 : SUCCESS in 2h 02m 14s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_PowerFlex_4.5/20c9ca3/DellEMC_PowerFlex_4.5/184 : SUCCESS in 1h 57m 41s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/34/916434/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/6d47f28/DellEMC_PowerFlex_4.x_NVMeTCP/120 : FAILURE in 2h 53m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"623144b18f867d7fd9714a65b4636e7a2b25b0db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0817060d12ea8444f3d3f0e13276fe7f7b69ba7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:14:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db4e45907144046929be69fa711c60f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e82b6a03d7a346ef9e1ab6d1c9ccddce : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7eb67d68e7eb4e5ba15aecd088761f6e : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd11f206f3c5cedadc21d089533386f0049fa04f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:01:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe291007f7764e4c90aab51651b5c556\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78176697061450e8223c9466b18d619 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e582f2f771bb4f219f2b295c66d89b95 : SUCCESS in 15m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc1cb1f5e4144995b69f7e9eca49ccb4 : SUCCESS in 10m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2e279c2f9bb4cb380f1beae5b7f3188 : FAILURE in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354104f8eafb40df968256e19edd8662 : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/79d6c4d1672a4f7c8a36debf5d357e38 : SUCCESS in 58m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28aa02fbfb9441bca91feeb4039b353b : SUCCESS in 1h 38m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0706de05564d4bdca90ff0d7bbe4901c : SUCCESS in 1h 02m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9fc5be814dce4c14b6ed01559abceeb5 : SUCCESS in 19m 02s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c8478a0aa42449a7a43f5819d3d67cbd : SUCCESS in 17m 49s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/942b9abd397a4966bf6f10c68c1f8acf : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/62f17d0e669246dda7b4d85fd09ebc31 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c606dc062a643b78e514071eae4966c : SUCCESS in 8m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/14aada1b8f97414f907f6d2c2e57e9f8 : SUCCESS in 7m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7839b296da914ee7be0aa07537005fc8 : SUCCESS in 40m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad20c7bdbb4c4dfe91ff3867aec7c9ba : SUCCESS in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1eeaf756eb7a4861bcee4f75767c0122 : SUCCESS in 1h 50m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/596ffc1058dd4f08bbc66832f645c8e8 : SUCCESS in 1h 59m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4513b12f4b1f46a8964c09c7071ba42e : SUCCESS in 1h 41m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ab9578b8ccfb4d558833e1ae3e3fd33c : SUCCESS in 1h 38m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5d71acbd8f044c19a883b72a5c5d3747 : SUCCESS in 42m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a6c4010474c4d72b6cfe34c97285de8 : SUCCESS in 1h 13m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/462880a1150e4b08bb0cfa4127537b01 : SUCCESS in 1h 27m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/82856497b7a74a28b5d1a0f85bbb8cf5 : SUCCESS in 32m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/af67fb1fd2ac426291055e36753116ba : SUCCESS in 1h 19m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d88061a8cab64de6ad0d5a00da1bd4ce : SUCCESS in 1h 18m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba727c48aa68480e95d90951ea6734fc : SUCCESS in 1h 38m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1949b6a1b4444f5a463264c5d451556 : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":3},{"id":"28d1fa285612dbaf7a91e0f87a8f6a6bdeb788b3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-13 17:40:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/3/upstream-check/cinder-tempest-plugin-ontap-fcp/56bf8bc/ : RETRY_LIMIT in 3m 16s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/60587fe/ : RETRY_LIMIT in 3m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/3/upstream-check/cinder-tempest-plugin-ontap-nfs/83ba5e7/ : RETRY_LIMIT in 4m 20s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f0cb9cf/ : RETRY_LIMIT in 3m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"190f26d2001d6f5266e44092b857555a135bfab0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-13 18:24:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916434 : FAILURE in 47m 44s","accounts_in_message":[],"_revision_number":3},{"id":"681d89174f7e8aa0657d19065188df3515ecbf06","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-13 19:00:29.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c7c2e99b0ae74708b002930d4614fed3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8cb640db27eb485a8c1fcd50ed7e4acf : FAILURE in 1h 25m 14s","accounts_in_message":[],"_revision_number":3},{"id":"a7b1f9af264335758bcda53118731e0089953766","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-13 20:18:33.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-916434-3/FC/ : FAILURE in 3h 53m 21s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-916434-3/iSCSI/ : FAILURE in 3h 56m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"db945c811ecd6b12347f30160067e6816e544596","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-13 22:33:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/916434/3/check/DellEMC_Unity_FC/4b7e69b/DellEMC_Unity_FC/1844 : FAILURE in 1h 53m 20s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/916434/3/check/DellEMC_Unity_iSCSI/74f54e8/DellEMC_Unity_iSCSI/1872 : FAILURE in 1h 43m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a8a27c24baf88bc4ef0331f52d03927a740a584b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-14 00:05:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/367/.","accounts_in_message":[],"_revision_number":3},{"id":"be17fd159dd30153f986df9261b530ca0ee9e6f4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-14 12:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/916434/3/fc : SUCCESS in 0h 35m 30s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/916434/3/iscsi : SUCCESS in 0h 49m 08s","accounts_in_message":[],"_revision_number":3},{"id":"52cd2185a35e8b2ffc4cf053bb622b50fd97fbdb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-14 12:22:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/3/20240514163547/NEC-ISCSI : SUCCESS in 4h 13m 31s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/3/20240514163547/NEC-FC : SUCCESS in 4h 46m 31s","accounts_in_message":[],"_revision_number":3},{"id":"bb530a871a727d1e9afc7d950d3ada88ba82fd14","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-14 23:16:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/34/916434/3/check/DellEMC_PowerFlex_3.6/3452e73/DellEMC_PowerFlex_3.6/1326 : SUCCESS in 1h 38m 23s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/34/916434/3/check/DellEMC_PowerFlex_4.5/6ed89f5/DellEMC_PowerFlex_4.5/251 : SUCCESS in 2h 18m 24s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/34/916434/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/2477e11/DellEMC_PowerFlex_4.x_NVMeTCP/198 : FAILURE in 2h 54m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"135873b26ff55caca0e1511586a21d875b872b44","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-15 17:33:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/3/20240515202141/NEC-V-FC : SUCCESS in 3h 12m 30s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/3/20240515202141/NEC-V-ISCSI : SUCCESS in 6h 11m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4bdd05b029455e9185c5b8ee150ea3fb8f26c24c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"69d61295fb670ec7c9242160423351908f99eb5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"3b4ec7407fe555e114ac7ed3911cdea7a1e15ff3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:28:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df81922171f742b1bf5d111ee8e65f33\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dbc8de27fcb42d9ba22458273490094 : SUCCESS in 16m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/56bc971fb1fc4b458ccddbac87af637c : SUCCESS in 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bcc7ff50b51920156c0ed7120250b223817f18cd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 11:45:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916434 : FAILURE in 23m 21s","accounts_in_message":[],"_revision_number":5},{"id":"8f94e54937f7cd5fc4606f81902d18893e9af890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:13:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa67da7bee3b43e996652ca8257bfe08\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9f9aff143c4e0a9b76f0c038b8dc81 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d7fee9928f44eaeac98ecb1aa1bc2f1 : SUCCESS in 12m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/835c92c66136414dacc5ebc6fe6e99a8 : SUCCESS in 10m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e9894bba0ce4efea5980543e71ff5be : FAILURE in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa772bc0bce5455aa1454a7c0fe5f774 : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/e00f60f1baa64ed6bb75144c9567c48c : SUCCESS in 47m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44fb23bb40854e11ab8567523060081f : SUCCESS in 1h 25m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd099b94b85e48019f4f6aeba29c8639 : SUCCESS in 1h 14m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/699b7bc3d3d643ba990f17f157ee4b91 : SUCCESS in 17m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e05a5392cc0046428220b6cd23df8d5d : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d72f4b2811c54f94b721d567bf710fa2 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f65e1adff1d04ad9b4d6cd3ef42ee5a7 : SUCCESS in 6m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/abe4c2a3564248b89da51ed0f99ee827 : SUCCESS in 7m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f8d01f60e0e94eb39ebab479eb4ba916 : FAILURE in 42m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/812f167b2c3244d09511060a0410f20a : FAILURE in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0e2db0c60f4b4846998d8460787a5832 : SUCCESS in 1h 55m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2176cbb2f77e4860af8081601ffd0229 : FAILURE in 1h 58m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8fce329d99d242ed8a7bd80c42b76cde : SUCCESS in 1h 41m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4188c16cd05147ab9780383925244e45 : SUCCESS in 1h 58m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/52c49fd82f6d42d18ce3e8fa8deb7dc1 : SUCCESS in 33m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e6747c85db64b93881226c12f0ef073 : SUCCESS in 1h 10m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a97e4f5faba456eb8b481ab876d4c93 : SUCCESS in 1h 33m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8fbc12347aaf45ab958229fc31a26bd3 : SUCCESS in 36m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f0b9fde1d3c543d8aa88fc5244700c8a : SUCCESS in 1h 21m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a7c0f0da8c964fb8ba43a84a71fcefce : SUCCESS in 1h 05m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/47d4ad3bb59b4d1fbe56edbcbbe04b3a : SUCCESS in 1h 37m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/153fad6fe2334cc98e2d46673d9e93ad : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":5},{"id":"a2441e8aa58e03f3795498244e0f19bfd2e76687","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-20 12:41:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/5/upstream-check/cinder-tempest-plugin-ontap-fcp/110747a/ : RETRY_LIMIT in 3m 44s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/23516d9/ : RETRY_LIMIT in 3m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/5/upstream-check/cinder-tempest-plugin-ontap-nfs/43ffb1b/ : RETRY_LIMIT in 5m 15s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/916434/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/34b2c7c/ : RETRY_LIMIT in 4m 49s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"70747c60512543ac341f5fcc51a7c388ff4926db","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 12:55:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916434 : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":5},{"id":"ac1a9c4900c2011243338f2c46a8c121c826b775","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 13:51:36.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5457d9f70e87445b84864ab2dcf11a9f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2024fcfc516f4420be3ebec302310c3e : POST_FAILURE in 21m 08s","accounts_in_message":[],"_revision_number":5},{"id":"baa52abb2d903ad5aacf4336218472617b365650","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-20 16:25:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/427/.","accounts_in_message":[],"_revision_number":5},{"id":"bf67c8f46cfc95ab1c1399ca2e596383b327fe14","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-20 21:10:02.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/916434/5/check/DellEMC_Unity_FC/998e658/DellEMC_Unity_FC/1904 : FAILURE in 2h 15m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/916434/5/check/DellEMC_Unity_iSCSI/2510fd6/DellEMC_Unity_iSCSI/1932 : FAILURE in 2h 02m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"7818d2dc12631ae4627cc06d5aa8aaf7a2c586e6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-20 22:18:50.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-916434-4/FC/ : FAILURE in 9h 48m 36s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-34-916434-4/iSCSI/ : SUCCESS in 5h 10m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"13e18b1e3258b04e19c557c5f219efa7d4549e81","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-21 00:17:19.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/34/916434/5/check/DellEMC_PowerStore_iSCSI/9b35580/DellEMC_PowerStore_iSCSI/1347 : SUCCESS in 1h 26m 04s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/34/916434/5/check/DellEMC_PowerStore_FC/25bf841/DellEMC_PowerStore_FC/1530 : SUCCESS in 1h 05m 08s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/34/916434/5/check/DellEMC_PowerStore_NFS/cb4dd21/DellEMC_PowerStore_NFS/1041 : SUCCESS in 1h 05m 09s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/34/916434/5/check/DellEMC_PowerStore_NVMeTCP/f8b1a6f/DellEMC_PowerStore_NVMeTCP/895 : SUCCESS in 1h 35m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"769865570db48d8f039ea9c9defd0e5cf98e0479","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-21 15:17:53.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/5/20240521165021/NEC-ISCSI : FAILURE in 4h 46m 28s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/5/20240521165021/NEC-FC : SUCCESS in 7h 27m 29s","accounts_in_message":[],"_revision_number":5},{"id":"61bbb1c23e3bc324851590161b121077c8f7577c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-22 01:37:05.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/5/20240522024435/NEC-V-ISCSI : FAILURE in 7h 20m 36s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/916434/5/20240522024435/NEC-V-FC : FAILURE in 7h 52m 26s","accounts_in_message":[],"_revision_number":5},{"id":"1f9da8f101b923a7875200b1f3206465b49088f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"42cdc7d563b689a724de361d4c9dd8d33c7f16ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:08.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/582/","accounts_in_message":[],"_revision_number":6},{"id":"b0d47b0f560fbb0f34c8c1991eac81c054a2eadc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:09.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916434/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916434/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"7dc2eb909e8967fb15139a600bfacd2ad80e854e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 15:57:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99835c0453ba44ce8524f191c3d163ec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02f408c654b64332a58a6aacd073a432 : FAILURE in 21m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3479263c282c425c805e2b6629b2c0df : SUCCESS in 21m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"55797c67f012516e473464c4616006bce22c5313","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 16:51:46.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cda38f5e87b54c7497bc8e2e44b247b4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ad20913792c43b9a9f2ec85d73c7010 : SUCCESS in 44m 27s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33dbb29d5061418ab16d5030d417a688 : FAILURE in 55m 24s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7d935260a824cdb8fd92184045f7c84 : FAILURE in 1h 31m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e4ed1dbacd3441dea6f4d9c6e1a9c2c2 : SUCCESS in 39m 54s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"b49830d1e8490d29ba8dba1de6dd90b485c430c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:57:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a21984aa9954617851eab0680f22b35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e2eee5b2be4b038a63e7e2754f4737 : SUCCESS in 15m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/703819ae378f412cb56116176f9a0317 : SUCCESS in 13m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5336dd1819b4dd48aaf64090ab22fba : SUCCESS in 11m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69c9b8ea078e4dbcaef6d4392c501575 : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7286fcab4040430bb8c3a456b43da2b8 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/58844836ed7e4ab19e8de44cd4bd921f : SUCCESS in 1h 04m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48883713bde44c119d8cdfb9157b3a82 : SUCCESS in 1h 57m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22fb87390f29402d9ed37f5f2a2b23fb : SUCCESS in 30m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d6cec221d5640b0bf9831ecbfd8030d : SUCCESS in 17m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca324cbeccdb4fefb18e5ec57bf11bf3 : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f1a6aeac51945e5bfed71ad26962028 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f67a008796ae4e66b82cb251991aec5a : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/492f9f07f99e4c3cb9c7bc1774c086b8 : SUCCESS in 18m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a0730b68fe62493982d789fda3f4470b : FAILURE in 42m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4664912b689d4ecb90c08831f18539fc : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/954880e326e045779e600ac26753bc54 : SUCCESS in 2h 04m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/76c9e9ce237c42a3a7e9aba026504fd1 : SUCCESS in 2h 24m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/933c11ee6b0c44d59ee92e49da0a9cc6 : FAILURE in 2h 20m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/33a61a241c8140cfb47dfbe15227c7ed : SUCCESS in 2h 06m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6e6a3c58ec9e46ed94b439ccf531573d : SUCCESS in 32m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/33ab301d8fc842b59a84952fd80865ef : SUCCESS in 1h 36m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2561dd94760e4a03a42a2cddf27b47a4 : SUCCESS in 1h 19m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a66172ba08804238a88cf4051f53eed8 : SUCCESS in 42m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/38e9b0dc6bbd47ffba30e23ba8e35ec4 : SUCCESS in 1h 47m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4c2bf88e767f41a2b33fd8202c813752 : SUCCESS in 1h 46m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a6a7372bbde64bd7b1f03a900705ef3d : SUCCESS in 2h 16m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db840cc9cb08431886121d9bab890576 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":6},{"id":"3b1c9a349de0e940edbba3baa01edb7165b884ed","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-18 18:28:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916434 : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":6},{"id":"9429bf4509779e4ea6907e6dda83c3d670b8eafe","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-18 20:24:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916434 : FAILURE in 27m 41s","accounts_in_message":[],"_revision_number":6},{"id":"c535a35d627cd14522e07fcc7589abb5464f342d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-18 22:35:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916434 : SUCCESS in 2h 57m 58s","accounts_in_message":[],"_revision_number":6},{"id":"4a6caefa67158719800bd111586def978fa5030f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-18 23:32:44.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c73d8ea21e7e428d8db3ebf377c05620 : FAILURE in 1h 21m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e4b958033c3b40bb8c5d527f1ffaa924 : SUCCESS in 1h 17m 23s","accounts_in_message":[],"_revision_number":6},{"id":"8453b9c11b8e1a3f916b131b42cfc83b6d49fd1e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-19 12:15:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/34/916434/6/20240919164616/NEC-V-ISCSI : SUCCESS in 4h 23m 51s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/34/916434/6/20240919164616/NEC-V-FC : SUCCESS in 4h 29m 07s","accounts_in_message":[],"_revision_number":6},{"id":"303473182b88e854064e7c0cf9fe47013fc9ffc5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 13:58:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_Unity_FC/df78de6/DellEMC_Unity_FC/2624 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_Unity_iSCSI/6e81b03/DellEMC_Unity_iSCSI/2633 : SUCCESS in 2h 29m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"6c9dbc03b20146a63101044ad8f31425c438bc2a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-19 18:01:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/557/.","accounts_in_message":[],"_revision_number":6},{"id":"861df7a8fd655712fabcaf34e164645369070892","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 00:13:51.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1145/.","accounts_in_message":[],"_revision_number":6},{"id":"5e42d84f3a701c0d192536caed6be02275901823","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-20 12:29:42.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_PowerFlex_3.6/88f230b/DellEMC_PowerFlex_3.6/2118 : SUCCESS in 2h 17m 29s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_PowerFlex_4.x/fbd9241/DellEMC_PowerFlex_4.x/1002 : FAILURE in 2h 22m 27s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/def2ae4/DellEMC_PowerFlex_4.x_NVMeTCP/942 : FAILURE in 5h 53m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"50a63da1c6c823a1e999b4c2eb156f0f5685f5d1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-20 17:09:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/34/916434/6/20240920202218/NEC-ISCSI : SUCCESS in 5h 22m 41s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/34/916434/6/20240920202218/NEC-FC : SUCCESS in 5h 46m 40s","accounts_in_message":[],"_revision_number":6},{"id":"131b790a7ce5f8867ead53b97b72cc6f2bc550aa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-24 16:39:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5aa8fc6d4c32e3cf3d54d3c172ade078b8951723","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 17:29:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f3de6ba7f954a74906eca6d4193123d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3cf2c7177004656a3aefe99f80d423f : FAILURE in 27m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/18d2398938914f95aa7f7b4fc6bd42f9 : FAILURE in 26m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"24d532f3b2443ead20b6c18a908c9f70027c7b1d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-24 18:49:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/789b1581f4924028b7a2552b403d3510 : FAILURE in 1h 11m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b4249b17629c44bab0efd80f026dacff : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":6},{"id":"e526265173fed3337a67ad9f3d7f8421b89e1895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:34:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad87287860c4a5ab788b0d3bc4ac330\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce7a969fe9a4796b387e1c3126c7b51 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e21a461688bd4cd7b326e81aef0dadb3 : SUCCESS in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18ff3bee1bc64b319583de3ef9b43d82 : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55df4daa5a4492e913d75489fbf99ea : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/d024c88ed75541759bc00af52bfdd239 : SUCCESS in 1h 01m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44296e04e2fb4aaf8e9e6338e30090bd : SUCCESS in 56m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ef0122c945f4771a22d682d06d20ca2 : SUCCESS in 1h 31m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb56b1cc16254378b052d28425663ac2 : SUCCESS in 55m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d9c835a1a4841b287649bb46d0ad3ee : SUCCESS in 16m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4ce5256bdef04a4ea1c7c1b5912ddc66 : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2dbf599d3a98445bbe2747ab5d5e9e76 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b31a9a6df254821baa5a451d21e690a : SUCCESS in 4m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3110d848c0a74282a99921f45860e501 : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4db8948b637a49d0b504d7395587fefa : FAILURE in 32m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c60985e11db94334ac1692a69e700024 : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e7284919ada4915bb0903e79c7b4b89 : SUCCESS in 1h 42m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0dfb24e118f64d0e8ee3265f2e0ebb78 : FAILURE in 2h 48m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/544e0eaa1fdc4fb5ad91c9e1922f142f : SUCCESS in 2h 10m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9f2a3e72fe884142a51555ff9e54fa32 : SUCCESS in 1h 45m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/867b1412301b455baa81fb2aa42da277 : SUCCESS in 28m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8d483c91ac5942609b015c6ae3a880e3 : SUCCESS in 1h 10m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/af0f116cd38245598980682c23a8fd5f : SUCCESS in 1h 57m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6bceef29c01e4fc884e78d2610ebcd11 : SUCCESS in 31m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d927eae6dc964a6e838e2da37955ddd4 : SUCCESS in 1h 26m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/782290bbbb8d401ca8ace0ba1204404c : SUCCESS in 1h 02m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/101e42dd44694876be7dde7e0b737c24 : SUCCESS in 2h 02m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40fbced7fb43420dba12b91f9ba7d54e : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":6},{"id":"7dcf25d40290cdfbd9ce0318f2e6964fbb193c2c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-24 19:52:33.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/1813e3e686f2472ebf678559933b8055 : RETRY_LIMIT in 5m 06s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/3c97515d46cd4e49ad51627e315ad84f : RETRY_LIMIT in 6m 01s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"b0760db282e39e7f95a8e2cd252efd7f3c40611a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-25 00:56:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/916434-6/Seagate-iSCSI/ : SUCCESS in 2:20:42\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/916434-6/Seagate-FC/ : FAILURE in 2:24:17","accounts_in_message":[],"_revision_number":6},{"id":"04fe26157b339b3999f63f2fb0e6f5efc60349e5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-26 11:07:42.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_Unity_FC/06b3fbb/DellEMC_Unity_FC/2792 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_Unity_iSCSI/bb8ddcd/DellEMC_Unity_iSCSI/2777 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"03633bfbafd7800f066ba50a0e22d46c51f71d22","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-26 12:25:35.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_PowerFlex_3.6/df7026c/DellEMC_PowerFlex_3.6/2296 : SUCCESS in 2h 13m 32s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_PowerFlex_4.x/f092cb1/DellEMC_PowerFlex_4.x/1178 : SUCCESS in 1h 49m 35s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/34/916434/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/a6cb9ef/DellEMC_PowerFlex_4.x_NVMeTCP/1092 : FAILURE in 5h 54m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"fac855093a5dd62fd3a18a52106a67bde1e5a47d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-21 17:02:34.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"993b11900d4340d99c32a9b3ba790c9714be3669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e5e249164ae130495a9a8e8fe331b658df7a5de8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:49:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"fd2156610783640959a1dd2224d3462c6507e78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"a472dee1df5a1368aa09d34d94704e59888d93a2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:05:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"f9b80dc87ba9abc536587739d41929c2fece3062","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:19:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d0568f217e44749b81379c820c1147c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67d92aa481aa4307bc1c712c0f0128ee : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd75605e53ed437a8f5f054d9fabc1b5 : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85c430e3f83744699850739e1709b945 : SUCCESS in 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"714e5aad64774a006babe8117be63c208e9ed83a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 13:46:03.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia3104b702839688852fc535f01b29018bdf3c9a5 .","accounts_in_message":[],"_revision_number":7},{"id":"dfb308c3286de3290749504230162b6bde91fbb3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 15:22:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2247/.","accounts_in_message":[],"_revision_number":7},{"id":"a1bd0d24c78c0858ebb45317a25c916cf2d0f82e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 15:36:54.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia3104b702839688852fc535f01b29018bdf3c9a5 .","accounts_in_message":[],"_revision_number":8},{"id":"48365093d4fe5f9c37c6d8687db925de7812b984","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 15:59:21.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a9da700a1fe54c0a96f7e6a72b9317ec : SUCCESS in 44m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b2898522bb2c4572af3c4686b7c312d0 : FAILURE in 1h 02m 12s","accounts_in_message":[],"_revision_number":8},{"id":"7ddd83e7dfd6ed0ca82251f76f1004686a14cf45","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 16:10:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2133/.","accounts_in_message":[],"_revision_number":7},{"id":"a483500bfe9ef0dc9816c0b8fa696d6361c4a47e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 16:11:33.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916434 : FAILURE in 1h 13m 31s","accounts_in_message":[],"_revision_number":8},{"id":"67dfd8caa834f60ec65d1f5cd9696270fb6f8731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 17:33:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a7d3b288e71496ab6ff2e97d26b8d85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb002a77a2c483da35a008bbc301e52 : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c1165693c3a4c1a8abd1e19f828b08e : SUCCESS in 11m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/790bec0d444445bb8d653536a5c0ddd0 : SUCCESS in 16m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0a1e7e87f0545f9a23b5a5520c26893 : SUCCESS in 15m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/8f1e57efc9e54c45b53f3d26f40acd39 : SUCCESS in 1h 01m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0fbd9bae017f4883860ab47256a6dc07 : SUCCESS in 1h 00m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/758c162792454a56a24a44372d475f4a : SUCCESS in 1h 29m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b17ba62b06b4d038466f6073379d7e6 : SUCCESS in 1h 04m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2e698522a2e405e99502dd07056485b : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fe686f1679c94189b355fcf180f876e3 : SUCCESS in 5m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e469b0d72f214c489036c29c0565f80b : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b27218a28684e92bada857c26c1a9a4 : SUCCESS in 5m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e51522f88b7147f8a4bb7bf4899d507b : SUCCESS in 8m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/198b7de2e166474393d3752e38c28c22 : FAILURE in 36m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9d67d536a11e426fb19ea3d8e3130f3c : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f6fc7f4679743c7a71641c8604bbee0 : SUCCESS in 1h 54m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8dd635b434a7440c9cbafe145f68a9f6 : FAILURE in 1h 27m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3efa38c7e40c437e99f6aef8d7962a59 : SUCCESS in 2h 01m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/60c0cc28c890444691584828dd26134c : SUCCESS in 1h 55m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/801903871d844a179b5abb10806b6507 : SUCCESS in 33m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/150aec29617245e5a3d7e1efb02f825c : FAILURE in 39m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a9f40ac736d4d139ee39af2861fd983 : FAILURE in 1h 40m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c8e5a1d739784383968b09ea8d53ec3a : SUCCESS in 42m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fe78853ee8d2488d88763b9a79b7c462 : SUCCESS in 2h 12m 30s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ed2500b24f4842beb56081fd80e9e95c : SUCCESS in 2h 13m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a8c249d3031495e947d4a2e0fca35f3 : FAILURE in 1h 54m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/981162dc75a84c55a6d90b661fd52668 : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":8},{"id":"f46a1e7001951f13c4702e6b9b55a0555615086d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 18:13:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"415137496de397683402bc7886de4e632e547631","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 18:54:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4976a35b890a4b9e8fe55fbec8b81c1c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99a093e7a0514211937a9615e517df65 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7dfbffaaac9a4d30abf7c19eda22a654 : SUCCESS in 22m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3d2693cc72f8451e8c5eebeb712f0d66 : SUCCESS in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b7fd4a33c268768bcb2f73148ef93367505008ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 21:05:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8275c0d6763a4024a81b02c66d246cc7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155ba2b6a0dd4ce78d03a114f529164a : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8dd7b03dcc4a4ab38ee3cf934dffbe46 : SUCCESS in 11m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24734e7d98dc4cacbc40ceb6201d0c9c : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4111356b79417d9ef4905e0818243d : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/2fd6706c17024198bfca7be883c877a9 : SUCCESS in 56m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca4b0b727d6e40d59a9ee49987d6330a : SUCCESS in 43m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dabc50107594476eaf0c9c868fa0abbd : SUCCESS in 1h 35m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b53b6a285d44d4c961a67b5d4ebb05d : SUCCESS in 1h 03m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aa5d496873e648c69ae336d29a9d1686 : FAILURE in 25m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64ca6954cbdc45bca0af834686b71cf6 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e7039ae1400e4237a86634c40687d260 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f97c11efce945d681a9825d9b9bbc91 : SUCCESS in 5m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5267568c3d00445097eb4a9e7d5c4907 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2694c9c6de594d09b88373618bc63b91 : FAILURE in 38m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4fb58805bafb4aabab9ebb3a170d5efb : SUCCESS in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/17cf35b8458b460eb1d96108eda8280e : SUCCESS in 1h 02m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f44f0aea7bd4eeaaebb8ef8d9fb2dcb : FAILURE in 2h 47m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6a84db0c961947fbb3b7d9713d847fd2 : SUCCESS in 1h 33m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/eb9b3b8de82342ca8d9071978595e89a : SUCCESS in 1h 27m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/64fc7f279c0e402789817cb891a6d897 : SUCCESS in 19m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f9abe7ee071742b7bee266e5d561ff18 : SUCCESS in 1h 10m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/116b3d6124964afe9c28cb60553ef2ec : SUCCESS in 1h 40m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/973489fe2a29492699610259b30d788d : SUCCESS in 30m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/492dc3f889914d9494882e9110b522a4 : SUCCESS in 2h 05m 22s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a5042ede91d54a1691923753174f92d8 : SUCCESS in 2h 09m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5d63b149f4944dfea86c9ff482ef819d : SUCCESS in 1h 01m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27238c93385d4af09ffb5261854b6f12 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":8},{"id":"7f036354309b6cd300b40b7e0aa5de58c836dbd5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-24 21:12:20.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_10_2025_21_05_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"ec0566cde2df9ce78a60f02a7e33bf5b14e9b0d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 08:24:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2265/.","accounts_in_message":[],"_revision_number":8},{"id":"3bcf9035e2315034a6c6e1ab60c4e1d4574cac87","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 08:40:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2151/.","accounts_in_message":[],"_revision_number":8},{"id":"51b680ea8e5883c586c66c96c9f74436b00da54e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-25 13:10:17.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/06fb029bdca9440b928e2c2938cea31b : SUCCESS in 44m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fa6ac26b66ff4a48a941652e0bfea6cd : FAILURE in 56m 58s","accounts_in_message":[],"_revision_number":8},{"id":"2c01faf592e5ee9fdc91255ae6719b582a7fbd6b","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 14:38:26.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"8090ffed9d7f7f5d605bcd864e29b4aac8ff350c","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-10-30 14:38:42.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916789 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"da721bdde50a800fe8ba64db44b8451e57acb438","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-10-30 17:45:29.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916789 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"4fa03df934025e1ebfe04bc1ec0848043162a2d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:08:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8a039de3d8df61e94d4a5fab2138eca20ae69363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:16:37.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93137edee7444565ad65beaf33375128\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75094ddee9ae4444aa77024539b1f414 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64431f03d57149f9825c8cb14b126f8c : SUCCESS in 12m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83747d4cbd6346919ea1453a426a9eb8 : SUCCESS in 15m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d29649e5de52491fb42504f577346111 : SUCCESS in 14m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/b8ce85cdc9ed40b18cc02203b6339f79 : SUCCESS in 1h 01m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f7d1ac8272e403297292105e198557f : SUCCESS in 42m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6db3970bd7d24d19bcde26e8f1d4f65c : SUCCESS in 1h 19m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79349814dc8a42049826d8b7ae3d4563 : SUCCESS in 32m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/acc21c1cd12a4b9fa6417d2cf4bbc201 : SUCCESS in 1h 14m 59s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d405cfe57b944efca8c854dc6d76cd75 : SUCCESS in 1h 05m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41cccd9dd8594bf1aa4fc63fac7284f0 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":8},{"id":"f561a880dcd689a39c646ce8b45bd350a01377f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:16:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f6ae4100fca8037e1be46995808849b9fe1cd365","revisions":{"609b3a416cd32030fb785d10e71f7e64f148ff5b":{"kind":"REWORK","_number":1,"created":"2024-04-19 14:51:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/916434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/916434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/916434/1"}}},"commit":{"parents":[{"commit":"b36142602e4a478e0e9349126ab2bee92f3f2d18","subject":"Merge \"Fix snapshot status is always backing-up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b36142602e4a478e0e9349126ab2bee92f3f2d18"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 09:44:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 14:43:06.000000000","tz":60},"subject":"api: Move rate limiting middleware to its own module","message":"api: Move rate limiting middleware to its own module\n\nThis thing could probably do with being deprecated, but until that\nhappens we don\u0027t want it living in the old \u0027cinder.api.v2\u0027 package.\n\nThis is pure code motion.\n\nChange-Id: Ia3104b702839688852fc535f01b29018bdf3c9a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/609b3a416cd32030fb785d10e71f7e64f148ff5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/609b3a416cd32030fb785d10e71f7e64f148ff5b"}]},"branch":"refs/heads/master"},"df3612e4fb7311b6496070b53a25ec81c3f066d1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/916434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/916434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/916434/2"}}},"commit":{"parents":[{"commit":"44f07694582e11e98554c198228feebb91bca1c1","subject":"Integrate \u0027pre-commit\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44f07694582e11e98554c198228feebb91bca1c1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 09:44:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:51:11.000000000","tz":60},"subject":"api: Move rate limiting middleware to its own module","message":"api: Move rate limiting middleware to its own module\n\nThis thing could probably do with being deprecated, but until that\nhappens we don\u0027t want it living in the old \u0027cinder.api.v2\u0027 package.\n\nThis is pure code motion.\n\nChange-Id: Ia3104b702839688852fc535f01b29018bdf3c9a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df3612e4fb7311b6496070b53a25ec81c3f066d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df3612e4fb7311b6496070b53a25ec81c3f066d1"}]},"branch":"refs/heads/master"},"68282fa5c26480505c185f71871ada5a770164d3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/916434/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/916434/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/916434/3"}}},"commit":{"parents":[{"commit":"e9613b1bb23e3a834449a4a5074d783c061ab739","subject":"Integrate \u0027pre-commit\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9613b1bb23e3a834449a4a5074d783c061ab739"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 09:44:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"api: Move rate limiting middleware to its own module","message":"api: Move rate limiting middleware to its own module\n\nThis thing could probably do with being deprecated, but until that\nhappens we don\u0027t want it living in the old \u0027cinder.api.v2\u0027 package.\n\nThis is pure code motion.\n\nChange-Id: Ia3104b702839688852fc535f01b29018bdf3c9a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68282fa5c26480505c185f71871ada5a770164d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68282fa5c26480505c185f71871ada5a770164d3"}]},"branch":"refs/heads/master"},"1a60040c070cb729d763374deb6f6a62e8c0e260":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/916434/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/916434/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/916434/4"}}},"commit":{"parents":[{"commit":"51890c0d5162f407304c0e9e222ba20fac522224","subject":"Integrate \u0027pre-commit\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51890c0d5162f407304c0e9e222ba20fac522224"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 09:44:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 09:58:47.000000000","tz":60},"subject":"api: Move rate limiting middleware to its own module","message":"api: Move rate limiting middleware to its own module\n\nThis thing could probably do with being deprecated, but until that\nhappens we don\u0027t want it living in the old \u0027cinder.api.v2\u0027 package.\n\nThis is pure code motion.\n\nChange-Id: Ia3104b702839688852fc535f01b29018bdf3c9a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a60040c070cb729d763374deb6f6a62e8c0e260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a60040c070cb729d763374deb6f6a62e8c0e260"}]},"branch":"refs/heads/master"},"f8e875db6048eef6db73576e4303bd4a4f40c324":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/916434/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/916434/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/916434/5"}}},"commit":{"parents":[{"commit":"33204d6f37b6d433099f0c3d98bc729c912a1e25","subject":"Integrate \u0027pre-commit\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33204d6f37b6d433099f0c3d98bc729c912a1e25"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 09:44:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"api: Move rate limiting middleware to its own module","message":"api: Move rate limiting middleware to its own module\n\nThis thing could probably do with being deprecated, but until that\nhappens we don\u0027t want it living in the old \u0027cinder.api.v2\u0027 package.\n\nThis is pure code motion.\n\nChange-Id: Ia3104b702839688852fc535f01b29018bdf3c9a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e875db6048eef6db73576e4303bd4a4f40c324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e875db6048eef6db73576e4303bd4a4f40c324"}]},"branch":"refs/heads/master"},"e57fc8cdf36522c2e1864cf2ffed7334f2bf2549":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/916434/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/916434/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/916434/6"}}},"commit":{"parents":[{"commit":"841fad272c17fbf502e2c5d888d5914d1b65dd2a","subject":"Integrate \u0027pre-commit\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/841fad272c17fbf502e2c5d888d5914d1b65dd2a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 09:44:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 12:52:28.000000000","tz":60},"subject":"api: Move rate limiting middleware to its own module","message":"api: Move rate limiting middleware to its own module\n\nThis thing could probably do with being deprecated, but until that\nhappens we don\u0027t want it living in the old \u0027cinder.api.v2\u0027 package.\n\nThis is pure code motion.\n\nChange-Id: Ia3104b702839688852fc535f01b29018bdf3c9a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e57fc8cdf36522c2e1864cf2ffed7334f2bf2549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e57fc8cdf36522c2e1864cf2ffed7334f2bf2549"}]},"branch":"refs/heads/master"},"b6bc2b4c06f6f356b2dc865871a41a4251a0ce24":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/916434/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/916434/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/916434/7"}}},"commit":{"parents":[{"commit":"60d30a37f7d0b8f717baa38f128cc6555e8ea4b1","subject":"Integrate \u0027pre-commit\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60d30a37f7d0b8f717baa38f128cc6555e8ea4b1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 09:44:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:44:14.000000000","tz":60},"subject":"api: Move rate limiting middleware to its own module","message":"api: Move rate limiting middleware to its own module\n\nThis thing could probably do with being deprecated, but until that\nhappens we don\u0027t want it living in the old \u0027cinder.api.v2\u0027 package.\n\nThis is pure code motion.\n\nChange-Id: Ia3104b702839688852fc535f01b29018bdf3c9a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6bc2b4c06f6f356b2dc865871a41a4251a0ce24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6bc2b4c06f6f356b2dc865871a41a4251a0ce24"}]},"branch":"refs/heads/master"},"f6ae4100fca8037e1be46995808849b9fe1cd365":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/916434/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/916434/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/916434/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/916434/8"}}},"commit":{"parents":[{"commit":"41eee5c2191ba6cb3af30ef2d2a93d1287df5ad7","subject":"Integrate \u0027pre-commit\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41eee5c2191ba6cb3af30ef2d2a93d1287df5ad7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 09:44:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:53:11.000000000","tz":60},"subject":"api: Move rate limiting middleware to its own module","message":"api: Move rate limiting middleware to its own module\n\nThis thing could probably do with being deprecated, but until that\nhappens we don\u0027t want it living in the old \u0027cinder.api.v2\u0027 package.\n\nThis is pure code motion.\n\nChange-Id: Ia3104b702839688852fc535f01b29018bdf3c9a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6ae4100fca8037e1be46995808849b9fe1cd365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6ae4100fca8037e1be46995808849b9fe1cd365"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
