)]}'
{"id":"openstack%2Fcinder~916435","triplet_id":"openstack%2Fcinder~master~I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-30 20:16:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544","subject":"api: Remove v2 limits API controller","status":"MERGED","created":"2024-04-19 14:51:42.000000000","updated":"2025-10-30 20:16:45.000000000","submitted":"2025-10-30 20:16:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916435-openapi","meta_rev_id":"efac2b4c4e2ab9dac068fd2131444b7b03756d8a","_number":916435,"virtual_id_number":916435,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-30 20:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-24 12:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-10-30 14:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-10-30 17:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 15:10:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 17:38:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 19:56:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 20:54:38.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-20 00:58:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-20 03:23:32.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-20 07:54:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-22 15:19:43.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-23 17:00:29.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-23 23:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-05-02 01:52:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 00:09:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-13 17:27:24.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-16 03:56:16.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-05-20 12:13:26.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-21 01:50:47.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-09-18 14:15:14.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 17:18:43.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-18 18:43:39.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-19 00:51:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-21 17:05:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-02-21 17:11:44.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-02-24 12:53:31.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-02-24 12:53:39.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-02-24 12:54:20.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-02-24 12:54:27.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-02-24 12:57:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-10-24 13:44:57.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-30 14:37:32.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-10-30 17:45:30.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"06d9605530488fabba1e67b377a80cf027e87760","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 14:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93ab0a356afa18126b9d932229a2d4ae5b6ae184","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:10:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7e84a634e224761baa56d43b34f245e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7065f19c61c344fca4b494065b575061 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6b9ed46ae2444717885ea8b5f10d2346 : SUCCESS in 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcda71725f03a0bf2536812723825700aabbce63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:38:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11fa1e6702c54db88512804a9243992a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e18710f65fc745e8a4ba6a1a1241a9af : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc83c60891364d7881a368071c89eb36 : SUCCESS in 11m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/069e53d271c04ce6903d9436aa837d3e : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2082dcdb39e54c45a66eea6e69e2d184 : FAILURE in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9fc1a90b5bd4be0abb1dfad71b9e8d7 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/d642a95704d245a99ad2c5ba5c5923d9 : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d58003c18ee14ffeb3ef3172851e7920 : SUCCESS in 1h 27m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e865122dc7a49fcabb2d4591d2ae7e0 : SUCCESS in 1h 06m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/20b0424d9d8744e49cb3b06ab3f0a059 : SUCCESS in 16m 15s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/185c7a9272694e3e8fda0672d0ae7ba0 : SUCCESS in 13m 40s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c56806e47f7c4608867f69a497e1d597 : SUCCESS in 6m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98c4685491904e6b809c37b3dfcdd66c : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/189ed2c17ee54f39bbbbefe35018fb5f : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bad4a9a2ade943cfa1d7ee809422d296 : SUCCESS in 6m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f41613b9e7349d9994dc81c5c032425 : SUCCESS in 57m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bff4ab6ac1fd4ed6911018b90704e70f : SUCCESS in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/981bf5a4063e45088750e5eef1cd87f7 : SUCCESS in 2h 11m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aa7a16c9b60a4b8eb5c0c7d36f94ea51 : FAILURE in 20m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/20e4c65b812c4b54b738a9f0a6b90ead : SUCCESS in 1h 47m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9d5d226024ca43b88a2bd17b0420a426 : SUCCESS in 2h 05m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b331d9425842447d937093397dfe35a1 : SUCCESS in 40m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/41fc703ebee8410b890471463edcaec4 : SUCCESS in 1h 14m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f7ed79b726654a6f8617c31387dd4d6c : SUCCESS in 1h 33m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c8454378b41494e895dd3f4ecfffdd6 : SUCCESS in 51m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/20d1dcb3d91a4d69a1c9ece276d11b9a : SUCCESS in 1h 30m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e55698aa24ee4318a4b7f723ebdd56d8 : FAILURE in 1h 20m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0eed8f0714746a4be67d104fdc2d9a8 : FAILURE in 1h 30m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67833db6c2ce432299857e8c4450fd37 : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1324cf8e82af296aad71275d632fac9b73153232","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-19 19:56:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916435 : FAILURE in 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"414e3d30485f9d9985684f50e4fdf2e76169734b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-19 20:54:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/35/916435/1/ : SUCCESS in 1h 07m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5275724039dd462b7757e353392f4e3d83988f82","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-20 00:58:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/35/916435/1/check/DellEMC_Unity_FC/3057566/DellEMC_Unity_FC/1707 : SUCCESS in 1h 25m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/35/916435/1/check/DellEMC_Unity_iSCSI/cdb442c/DellEMC_Unity_iSCSI/1737 : SUCCESS in 1h 42m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5f9d9dee602cb18a452408c5d2e1dc1092d60b5e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-20 03:23:32.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-35-916435-1/FC/ : FAILURE in 10h 40m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-35-916435-1/iSCSI/ : SUCCESS in 7h 39m 52s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"185f97ae716503304c39a25c513790224d3b899f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-20 07:54:52.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1ad846fb48694033b53b9c0da8722f5a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/95395674d47f4424ad3a28fa73b06767 : FAILURE in 1h 26m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2c923b362009a07023f4f835cae1dabf83336737","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-22 15:19:43.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/248/.","accounts_in_message":[],"_revision_number":1},{"id":"46041736bb93c92da8ebe90bd059dc85d4e5dcea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 11:06:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc6dcb4b3c4eb07e4e27e4b0856377b3e25cdc11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 11:29:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc17cbf3840a465b8e2529300a9116fc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53adfbcfb8bb42fa8fe8190294d35baf : SUCCESS in 17m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ce761b8938346c8bd0e43abeec2a273 : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17176d9f6a0939719512f4fe4e25601f07ba37fe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 12:41:28.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/0ff82f647c574873b16fb9dc0304b3ba\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2b36565d93354716b578b025e22cb5e6 : RETRY_LIMIT in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7f335511c4bfa367d73f4bd67a6cfb9ff786e2a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c5bb7cc77adf6b7bdc9ebf21ff5a5b1e4a1a7bf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:14:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a0473db24b84027b5d52271d5006861\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c60e01d11a44ba1a1a8c42a3d03ac31 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d39e1a3396404b2a86220b9c78a3d7ff : SUCCESS in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4daa28137f3e5025c3a17ccb2f7d9c742307a27","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 13:25:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916435 : FAILURE in 12m 48s","accounts_in_message":[],"_revision_number":2},{"id":"95cc8b2941623e5b78d7aaec3278874a43becc28","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 14:01:47.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"3cd23745037db72ab0387fa198acf3c06dba2884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:26:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7905d27ca0f447679252666dfd6fee05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f9b9b5622e4759960f4c421256e2ed : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90557dce593743958f362aa08baf882a : SUCCESS in 12m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a74440cbc36f47f5b54cc7eb024d8551 : SUCCESS in 10m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7062b395855a46bca0d3417c5af4010f : FAILURE in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c67fd7f28549b2a8b4f589ef35793a : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/142ec80a95754728976fceafe4deaa81 : SUCCESS in 1h 01m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3f29016b37d4fcea1a75be06192f847 : SUCCESS in 1h 28m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fb04f44d3a04e6091327a909c7135c8 : SUCCESS in 41m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1382dbb9ebb541e5ba588084bbbeb90e : SUCCESS in 16m 30s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c164a8602c4b4ab899c2d452ee357013 : SUCCESS in 13m 35s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ef526ced947b4617b561732c47f91005 : SUCCESS in 7m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f2046f5c3254bfb9b3ce853ffbea4bd : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc4e33c5fcf444f38f12398640738309 : SUCCESS in 6m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3877c8c9c66140fe991903013f5f6ec6 : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0ca325844eb643578c6d1f043042e579 : SUCCESS in 45m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/adfbd190be25452abf2c5d0f6719c444 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f58b4a8962d4f5ebcd4b8f32ed456ae : SUCCESS in 1h 57m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9b141ac6c7e249e08236008ff3a30c68 : FAILURE in 2h 27m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/25cf442096ec417f9fda741005c21289 : FAILURE in 2h 00m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dd978fc265fb4aa1b892b1f074e6660f : SUCCESS in 1h 36m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b8a72f38c0ed44a88821a04632cfc220 : SUCCESS in 28m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ca67d06928945b58815e6c165d5d145 : SUCCESS in 1h 12m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/39e2de9a34454366b00dfd8efb71c40e : FAILURE in 1h 44m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4b8c397f8397451c8e0d3308720c3cc6 : SUCCESS in 42m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f29c5b4d3c8e4b3aae0b6699c2729c01 : SUCCESS in 1h 22m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c82a88ccbce44438bde20053fd7d2faa : SUCCESS in 1h 33m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/133ffc066faf4f268dbebe5335ae86fe : SUCCESS in 1h 21m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a4c2383cd2148f0a3c92ed7a4b0f082 : SUCCESS in 1h 14m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b839c12b2a9433b59eb896d6d41d99df029807b6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-23 17:00:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/916435/2/check/fujitsu-eternus-dx-iscsi/25f6988 : FAILURE in 3h 36m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/916435/2/check/fujitsu-eternus-dx-fc/8a9be6d : FAILURE in 3h 43m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e69b3a9642f5cad322d1befec1bb8ea0e79b267b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-23 21:13:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/35/916435/2/check/DellEMC_Unity_FC/71625d8/DellEMC_Unity_FC/1742 : SUCCESS in 1h 21m 20s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/35/916435/2/check/DellEMC_Unity_iSCSI/9a01f0a/DellEMC_Unity_iSCSI/1772 : SUCCESS in 1h 49m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c6b3ae7e4ed0fa91c12467e92866eb927dc9e39a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-23 23:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/35/916435/2/fc : SUCCESS in 1h 17m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/35/916435/2/iscsi : SUCCESS in 1h 27m 52s","accounts_in_message":[],"_revision_number":2},{"id":"1f766245872b9fb03064c08f043cac4d4334c38b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-23 23:24:48.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-35-916435-2/FC/ : FAILURE in 9h 1m 4s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-35-916435-2/iSCSI/ : SUCCESS in 4h 20m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"9c07c788287e7b0ab6d86f6e6594f5d654b05992","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-24 13:14:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bc9557790b11bda675c2b3f4c330f8e1e38ad11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:43:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daba4b5cf89c4de7a78eff0e0a2bfcd6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d6c82ac6b2e4abdaddfb7962fedf4dc : FAILURE in 26m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c91f1b4ad04e40f5b02f5ab3ff0c61df : SUCCESS in 11m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a7b4f31202ea2d68b04206d01fc0bd3926431cc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-24 15:06:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/283/.","accounts_in_message":[],"_revision_number":2},{"id":"9cb214d861ead3e099a85c06ce0d6092f70402ea","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-24 15:08:26.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6fb4d02f8c134a229620d2fd38ca3326\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a80002549f6f4c03af54b3c778448659 : FAILURE in 1h 25m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e1d5e394ccb29e55df3ae38d59cc0270b7e3f02d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 15:37:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e6d0d1d6ce74414805622fc251d2bd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52753b269a784d7fac0e0637f767a4ba : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad1b48899a5a450d8f17e83abdbf7123 : SUCCESS in 11m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79c28eab74d343b5b3a9fb2c7bbed0ee : SUCCESS in 10m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d348ff81f62447468b48d08449133ecf : FAILURE in 12m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acdbb6f854064215a1d11ebad2650dfd : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/23b2f1a0afd24040a3cc5abbfa21cd5f : SUCCESS in 1h 10m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0d257a06caa4d10a4ab2a31f474bcd0 : SUCCESS in 1h 37m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0a3e872d8db44ecb59a9a33c8444208 : SUCCESS in 59m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/085187adc9014138ae2898b1c2769be0 : SUCCESS in 15m 30s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/aa9b25de531e45078d7950969b14863c : SUCCESS in 15m 00s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/195a0ca64c714b0fa968db7f9245cdba : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/96b8f2a73f0044b7a98f088d003630de : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/803d9b1bd83b4b2fbafcb79701b97de7 : SUCCESS in 6m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fc0e2b0e42624ec69e2adcebe366b369 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1dced0b1d6df4a49a597c83d83e0bf44 : SUCCESS in 47m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2248c078ca9745b9a2e8416dd3bc3877 : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b517e9f099ac47a582c766f7ad43555d : SUCCESS in 2h 03m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d2be2eaf300e4164b147b2abd840d065 : FAILURE in 2h 19m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5dd3ac0092be4db7b75499a2e488b52b : FAILURE in 1h 34m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/81239f6264894a29be58665e9ad80efb : SUCCESS in 1h 56m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d3f2321642d141a0ab239b4e630de27e : SUCCESS in 27m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b17e61dd84e64d16979fff52c9b82bcd : SUCCESS in 1h 15m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7c11f7b78eb74cd3acd87a378e01d49b : SUCCESS in 1h 34m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/561320edb6544e658260bbfa184ee52f : SUCCESS in 42m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/172ca5e195f64cd3a4bb56f6a6f31700 : SUCCESS in 1h 16m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/22c0c64b608e47c8b14b0f27f21255e3 : POST_FAILURE in 2h 10m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba7ca85a66d34bb181bdff3d2dc31507 : SUCCESS in 1h 41m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d217e7a3cb6940cd9d8cd15724c010d3 : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":2},{"id":"d6333acd44868c52ea318ed262a585b901e9bcbd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-26 16:02:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dcd352d428b0a7b430837bedfff346417133a9c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:33:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2165e75c9bea4e6aa00318f7d2599f55\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c57ca5a47d3a48a6961dcbdf3cf2a3b7 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/075e6dac168044c1b93a2c4d29902154 : SUCCESS in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49df0af9d60de167ee68ec0e93c4ec0746b3ff77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 18:41:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe15dbaafee149c19dfb5d0caadbd000\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24443cbbfa1740d1b52fb26b87778b34 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37ceb339912c408a950dc4a70f6535cb : SUCCESS in 11m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51c29e9af5e54c8794889a237b2214d2 : SUCCESS in 10m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56d6829b8c2a4c3983eecd518b8e0e8b : FAILURE in 12m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c699042535849ef9abbcaa21faf5cdf : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/925e4168ae3146e08f71338613b14103 : SUCCESS in 56m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ce621bb0be14ccba4d7409dcb402739 : SUCCESS in 1h 42m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a20d863fe8c84bffa6feb65eeac427bd : SUCCESS in 1h 06m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d05f90a29814f2a83706dbf21245394 : SUCCESS in 17m 57s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d21643e405a04e9cbe4b63604d116c35 : SUCCESS in 12m 33s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2afbc50d261c43ac8684dda00e012d01 : SUCCESS in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c09857a29a9f492fac0da2f4926d89ce : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/967763c45429430a99a76cfcd1405ee3 : SUCCESS in 6m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a4398dfb9a6e46a68294691808a7800b : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b31edbbb947e4c64b313645c0414840a : SUCCESS in 45m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aac666c4d58041bbb9034c86f5b4aafa : SUCCESS in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5529632b368a41f8adfaaddc6f8f35f2 : SUCCESS in 2h 06m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/56020587ceeb4421b0efd4d62be32d8c : SUCCESS in 2h 15m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94f8b2ad933742ee90251535a08382a1 : FAILURE in 1h 55m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/244786fd55b14646a32865ccec9c57b5 : SUCCESS in 1h 32m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e1a94bb02d534b98a5c1c09d3d91e1d9 : SUCCESS in 33m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f2174bbb6133462ebb25228220aab54a : SUCCESS in 1h 12m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f253e47d6e1340cdbe3656eb5a9d3dc7 : FAILURE in 1h 32m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da1b07e7a58c4bb5b3fe9dc43c7ac609 : SUCCESS in 35m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0a4fb680422d4a54aaca988934e1d2fd : SUCCESS in 1h 26m 25s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/33325fb6beb745778e84d2cda6ae4511 : SUCCESS in 1h 38m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9527a778abc0404388e2768a298cee6b : SUCCESS in 1h 37m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a72c505f4a143cc93a2318109dedd0a : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b4fe3dfcd478a57fbff40fd3937eab20b659799a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-27 01:00:25.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/05aeef66b3164e8ba7c0e5938ce1c0e5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e82563f57d414f6aaccd400c2f66b270 : FAILURE in 1h 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"acd658de98aad040d7901c621e78d27bf649fa80","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-02 01:52:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/2/20240502023720/NEC-FC : SUCCESS in 4h 45m 24s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/2/20240502023720/NEC-ISCSI : SUCCESS in 8h 15m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2b22199838fe6d9e8aa7bba205f83fab9303f867","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 00:09:11.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/35/916435/2/check/DellEMC_PowerFlex_3.6/99a791e/DellEMC_PowerFlex_3.6/1244 : SUCCESS in 1h 42m 49s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/35/916435/2/check/DellEMC_PowerFlex_4.5/20b172e/DellEMC_PowerFlex_4.5/185 : SUCCESS in 1h 45m 13s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/35/916435/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/ed541d3/DellEMC_PowerFlex_4.x_NVMeTCP/121 : FAILURE in 2h 54m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"0f116bd7955b21924d13b5e163e2240d5f06b051","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d5e9e5d660c0449475429f2c09aa0993e337493c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:14:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c13d89f5c19f4d7abb8632797c7e9206\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0357773bc6694fc4820bf4100fbe6c83 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba554c75c8dd44e9acc8f51e2560ce3e : FAILURE in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4843dc871a1e90b3bd2afb83a91d4e17fbedc204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:17:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d127f9a1ca464b87af6a74b1a33414\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e170755989da41a3a55cf4c26007a073 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c18c4fc7ac5546658d7a31ea3a852261 : SUCCESS in 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8aa306707a1048bf8c4046c2233a4810 : SUCCESS in 11m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85d88a07394b4b4b9b9643fc5c0a1ea8 : FAILURE in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af996ecee7d7490eaa392527b97f4093 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/9ee84a0f149d4d0fadff57480546696d : SUCCESS in 1h 04m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/590c0a6a860b4b7594dbdf3102de35ae : SUCCESS in 1h 20m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd935bccb9b544d1bf198fcd69695ad7 : SUCCESS in 57m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4522513bcb1748da996de64e78a6574f : SUCCESS in 17m 32s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e6efc85c31c5497d90cad5d528da03ff : SUCCESS in 14m 15s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e12c4d32102e41448df898718b79cacd : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fcc571f2758b40c2b50f13e1dba30689 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88a3356cad364a2da0b7140c73d31bf6 : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ff77ce16c8cb4a39b2b5b471c130dcfc : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/73e470ab8afc4b869a78c166e90a3746 : SUCCESS in 51m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6d13d57ab9644c0d92508bc3c70980cf : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff7e694d585540edbeac358eefd5f156 : SUCCESS in 1h 56m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/402057c237c14a279ec54dd42149b252 : FAILURE in 2h 18m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e58da51455e4e89a84ca422a349b664 : SUCCESS in 1h 41m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d42e3a3c5c7e4b4ca722cf38bf3b1500 : SUCCESS in 1h 53m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/16b1aed89f18407a84b1b2feee8b4c9c : SUCCESS in 39m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f390604c058d443eaf95237ecf699cad : SUCCESS in 1h 16m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/571fe74376da43cf801b28ccf1e1c838 : FAILURE in 1h 45m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a8cdb0810a3494f94d4053b191d7568 : SUCCESS in 47m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c735d36268143dd83844a30ede9ca74 : SUCCESS in 1h 21m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/61e2573eba5241549bf6ac357a41b839 : SUCCESS in 1h 28m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ba1034f6a664848b329b3b6b9933f1d : SUCCESS in 1h 21m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69096e986b0b41dbaac64e9b11808f5a : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":3},{"id":"32a02efb0fc3ec8e968137cc4c915bb2077e554f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-13 17:27:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/916435/3/upstream-check/cinder-tempest-plugin-ontap-fcp/996c72b/ : RETRY_LIMIT in 4m 06s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/916435/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/d94a99a/ : RETRY_LIMIT in 3m 53s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/916435/3/upstream-check/cinder-tempest-plugin-ontap-nfs/eade79d/ : RETRY_LIMIT in 4m 14s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/916435/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/a8bf3e8/ : RETRY_LIMIT in 4m 09s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0d9408bfa2b6d55487823aa486fc9feb62e753e8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-13 19:03:34.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916435 : FAILURE in 35m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a82612ccf0ce395a751238954c7f05742b08b292","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-13 20:20:23.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/30855e0a6ec24915b92573c884b4ddfe\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3d02d10e14d74fd298b352228e2d40c9 : FAILURE in 1h 24m 24s","accounts_in_message":[],"_revision_number":3},{"id":"039ff242a52b3e718ca153f053d438da22cc47d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-13 22:44:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/366/.","accounts_in_message":[],"_revision_number":3},{"id":"406105759fb5e1a74ece42879c76548c8a2d14ee","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-14 00:36:44.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/35/916435/3/check/DellEMC_Unity_FC/562b428/DellEMC_Unity_FC/1845 : FAILURE in 1h 47m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/35/916435/3/check/DellEMC_Unity_iSCSI/7a81b34/DellEMC_Unity_iSCSI/1873 : FAILURE in 2h 02m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"1ba39f1fc5db0fa95c4efe6f4d501966b3668064","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-14 14:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/35/916435/3/fc : SUCCESS in 0h 38m 34s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/35/916435/3/iscsi : SUCCESS in 0h 47m 22s","accounts_in_message":[],"_revision_number":3},{"id":"f1737206290df499b1669ab2168bbae6ea4450c9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-14 21:24:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/3/20240514212225/NEC-ISCSI : SUCCESS in 4h 14m 00s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/3/20240514212225/NEC-FC : SUCCESS in 9h 01m 49s","accounts_in_message":[],"_revision_number":3},{"id":"55d4e74b5da006c3101c6911c62ccbca05404a5f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 01:41:50.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/35/916435/3/check/DellEMC_PowerFlex_3.6/6290eb8/DellEMC_PowerFlex_3.6/1327 : SUCCESS in 1h 53m 24s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/35/916435/3/check/DellEMC_PowerFlex_4.5/19e6a62/DellEMC_PowerFlex_4.5/252 : SUCCESS in 2h 24m 51s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/35/916435/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/ef6a163/DellEMC_PowerFlex_4.x_NVMeTCP/199 : FAILURE in 2h 55m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"39295db4eaaf1e6dee23a4d4b1ba1a60c241fd9a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-16 03:56:16.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/3/20240516023330/NEC-V-ISCSI : SUCCESS in 3h 40m 39s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/3/20240516023330/NEC-V-FC : FAILURE in 10h 22m 42s","accounts_in_message":[],"_revision_number":3},{"id":"db5667778b0240c7050824c15e6908b20621d39e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d3cca649852c40fce9010503f97075ca9f865e43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"fa6a747756b903a7a3fa8610ddb84d2baff1d384","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:30:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62cea99a6d144db8bc621b6b4c956b03\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f36b609a1cce409ea276f746db90febf : SUCCESS in 19m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ffa4c4340f441bab74689c949d1ffd0 : SUCCESS in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"baf141389586d9d4b2d40891ad8430fd9668f1cd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 12:13:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916435 : FAILURE in 23m 22s","accounts_in_message":[],"_revision_number":5},{"id":"25f7b4a10f526c7bca72dc6e077f8d7d5330c2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:25:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b114698fec574bbc9c8a37f7ae2f6281\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fafc3c661d8440588008bdb71e51f5c : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57cef26dd46a416389464a2fccdd5725 : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab04e75db9474b4c85b142573c11e6bc : SUCCESS in 11m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/079d944487ac4baca587f0778c85bf21 : FAILURE in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6380230f99a24826b19f0bfa813eb6e7 : SUCCESS in 14m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/d962d46b00044f37bf90da26c1e3d567 : SUCCESS in 1h 01m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21248a44df1544bcbd3c32d9e0bd4a42 : SUCCESS in 1h 36m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/053fba9733d54674ad658334a2b49cef : SUCCESS in 59m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0f95d1b060944e14b162719e335b945b : SUCCESS in 16m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/16e8e6eee31b4610acde76ba0afc860c : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9aea756b5864b5d976cd6b0a5e9b56e : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7eccbf68d54c4fac9b7860f4fe5f90d1 : SUCCESS in 7m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6e16636252334d789501f54addef84e7 : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82e2de9e7ffb4e87bab9370afbcc962e : FAILURE in 43m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7552131842b940a2bd797fc52c5f9009 : FAILURE in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8d3c53d9d7b641eeba5480327cacaf56 : SUCCESS in 1h 46m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/617c4c9126b24542bcd2bf5d7d5fd4f3 : FAILURE in 2h 10m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d94c8dd2d7124c92add9f4d8949c1c24 : SUCCESS in 1h 39m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7ca1f013e2fb4bafad7a640c835c2c72 : SUCCESS in 1h 44m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6d64675c4d1549ec96d9bf65de83079f : SUCCESS in 36m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce0f11f035074659b93be61e7258bb97 : SUCCESS in 1h 12m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/75808afc96ae4265925e8a6dc4b35c7d : FAILURE in 1h 50m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4f741215599c40398c895f03f05d4e38 : SUCCESS in 49m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3d5d40030d914ba2a821e08fe55a9a47 : SUCCESS in 1h 30m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4e894f9fee6840d49763c440e8ea64b0 : SUCCESS in 1h 31m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a5d648c21ec4056a0d69daa62b50a12 : SUCCESS in 1h 50m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a34bf726f64dc987d618df7998e3e5 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":5},{"id":"86cd31f17f4d1e75fc55df3ba122005b9e076080","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-20 12:48:29.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/916435/5/upstream-check/cinder-tempest-plugin-ontap-fcp/918b784/ : RETRY_LIMIT in 5m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/916435/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/27f67ec/ : RETRY_LIMIT in 5m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/916435/5/upstream-check/cinder-tempest-plugin-ontap-nfs/896b266/ : RETRY_LIMIT in 4m 21s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/916435/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/47a11fb/ : RETRY_LIMIT in 4m 08s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"51f14020325dc48d4a9f7c630fa7d161fb9970f4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 13:33:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916435 : FAILURE in 34m 41s","accounts_in_message":[],"_revision_number":5},{"id":"92fd61e9fb21364fc15283813cb5be948cc5bd8f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 13:44:20.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5a239b905a4846829bae6a68aae2be69\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/fda52878c02741e7b307a3991d8c00ae : POST_FAILURE in 24m 36s","accounts_in_message":[],"_revision_number":5},{"id":"963a6357a8d36c6b9f36daf0ae6009203a93109f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-20 17:47:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/428/.","accounts_in_message":[],"_revision_number":5},{"id":"b29cbbdf1045d6982e29415e9a05ea1a9bee49c3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-20 22:43:41.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-35-916435-4/FC/ : FAILURE in 9h 59m 49s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-35-916435-4/iSCSI/ : SUCCESS in 4h 25m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"70195ed33f951e1cd1c7907d359c1d2cc1e1b4c5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-20 23:32:58.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/35/916435/5/check/DellEMC_Unity_FC/d57a489/DellEMC_Unity_FC/1905 : FAILURE in 1h 46m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/35/916435/5/check/DellEMC_Unity_iSCSI/e2fc151/DellEMC_Unity_iSCSI/1933 : FAILURE in 2h 22m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a3c4a7ca636f27c48ea3e0e98ecefc663e6773c7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-21 01:50:47.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/35/916435/5/check/DellEMC_PowerStore_iSCSI/9b3b126/DellEMC_PowerStore_iSCSI/1348 : SUCCESS in 1h 18m 34s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/35/916435/5/check/DellEMC_PowerStore_FC/46a20dc/DellEMC_PowerStore_FC/1531 : SUCCESS in 58m 22s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/35/916435/5/check/DellEMC_PowerStore_NFS/21ad55c/DellEMC_PowerStore_NFS/1042 : SUCCESS in 1h 14m 29s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/35/916435/5/check/DellEMC_PowerStore_NVMeTCP/53edcc2/DellEMC_PowerStore_NVMeTCP/896 : SUCCESS in 1h 33m 22s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"a7a5ebf2bd9372abe4d46b6320387599e3955732","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-21 22:14:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/5/20240522001756/NEC-FC : SUCCESS in 4h 52m 11s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/5/20240522001756/NEC-ISCSI : SUCCESS in 6h 56m 42s","accounts_in_message":[],"_revision_number":5},{"id":"b6e54ddead04ff2f756f147da597f4e7ba22e365","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-22 05:06:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/5/20240522103708/NEC-V-FC : SUCCESS in 3h 18m 58s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/916435/5/20240522103708/NEC-V-ISCSI : SUCCESS in 3h 29m 31s","accounts_in_message":[],"_revision_number":5},{"id":"095944a49f6d9d1f7ab2ff83b0bea602a6251eee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ff9929c9cce62bbe56b39985a26f54047cbe58e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:14.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/584/","accounts_in_message":[],"_revision_number":6},{"id":"699474578638a6e8a3971d61c6e13c7f6a5467e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:16.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916435/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916435/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"9ca7aa0cc1931828750264eab8fdc5bf153598e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:01:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f59b9d760e74b96bd2781f57f85538e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07a6b9047d144823a47ebede3cea9f45 : FAILURE in 26m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/32a77e851ad749729057bf4ef1d7787d : SUCCESS in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"75d1776d92912a58055dfd470f17944da7002aaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 17:07:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90b3a711e23642e0ae2061b5136ec78a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1589ed435c244269772a7aabba6dd41 : SUCCESS in 18m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6295318b06de47cfb473ffeb3d7111e5 : SUCCESS in 23m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5a6af838ab144539f8170235fad062b : SUCCESS in 11m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/364345bcdbaa4662afa8291a5d1955a5 : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46414dba491e498f891e9b739ba26682 : TIMED_OUT in 31m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/82e4231458ff4e30be8b5384cd8c93b9 : SUCCESS in 1h 02m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1233df0833994252aef247ca05a40bab : SUCCESS in 1h 57m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b96600e038794ed2ad4b0a6dfa8d6a27 : SUCCESS in 28m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d797cda4884a456090658d5a7f040fcb : SUCCESS in 30m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dc6101a6d843408e89ada76995364c0d : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ca8aede50567424aa5b0b166f76577b2 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1b38833aa82147f59b1a0066289d7fe7 : SUCCESS in 6m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/893a963f23e14ca28e4fbd74a4e5e585 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7fd7f3143cdb41dfae0233b2a5b527ba : FAILURE in 37m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5e30516a46aa435ab2e797dfc94ad202 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/558e04633a654987b85105b23192e5ff : SUCCESS in 1h 39m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/43f9493388584a55a28712d9f97fadc0 : FAILURE in 2h 43m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/730e77bbe5d34528ac4de21cde0a3c09 : SUCCESS in 1h 40m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bc6adf3e84384a3884eb35e2f63d4bbc : SUCCESS in 1h 49m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f401b85e54b749beb9b8e74983caa8b3 : SUCCESS in 52m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a6cd5e768a5e4cdda5d2490c7a4a4335 : SUCCESS in 1h 08m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c889eae75f734fa583e71996ad7d8a7f : SUCCESS in 1h 58m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b1fa87831578483ab9041c3784cd6366 : SUCCESS in 1h 07m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e342e3429bb44c2199d66cc1349f4a9a : SUCCESS in 1h 53m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0fe71bf6d14c40c49fb7fe19aba4e4d2 : POST_FAILURE in 2h 22m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/538847eb223143c99571da55d169659a : SUCCESS in 2h 07m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20ae0016e51c41a79714ab3a3d584320 : POST_FAILURE in 35m 26s","accounts_in_message":[],"_revision_number":6},{"id":"55b8b185bf323190bca408ecc758828878798448","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 17:18:43.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/529a05ebe1194cc6992a9b26e35d0230\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7f5ab78a8546453c9dec289eea7bb813 : SUCCESS in 45m 06s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03b3a78b252c46a2adba010145842e08 : NODE_FAILURE Node request 200-0000007248 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/293d4d6f95a048b3a64c949508969b78 : FAILURE in 1h 51m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90d3bd904b114ecba1172d3c7ded02f9 : SUCCESS in 45m 03s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmai... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"c4d2f9236210497a86722534cf7416a899876c72","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-18 18:43:39.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916435 : FAILURE in 13m 13s","accounts_in_message":[],"_revision_number":6},{"id":"f262c35a29b0811b71f478fb1af2ab0bf08e7155","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-18 20:48:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916435 : FAILURE in 19m 24s","accounts_in_message":[],"_revision_number":6},{"id":"5e13538b7030a22cdc68db4581e0c43c66ecbdde","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-18 22:58:54.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916435 : FAILURE in 19m 23s","accounts_in_message":[],"_revision_number":6},{"id":"8ebe4e9d4d3790a21d3e21416b4da58d15d47fd1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-19 00:51:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/da3bca06a18e4a0ebb8c1cd8b6f24d7b : SUCCESS in 1h 18m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d9762fd77d254e8a91c75a30d0b25f09 : FAILURE in 1h 18m 08s","accounts_in_message":[],"_revision_number":6},{"id":"f01a9c9f95a9c0b978730ab6feade00d1797d07d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 16:58:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/35/916435/6/check/DellEMC_Unity_FC/31cdc87/DellEMC_Unity_FC/2625 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/35/916435/6/check/DellEMC_Unity_iSCSI/4bb139f/DellEMC_Unity_iSCSI/2634 : SUCCESS in 2h 08m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"b3d6240df22c58a6bb014a11a8ff415ecafbdb0b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-19 18:50:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/558/.","accounts_in_message":[],"_revision_number":6},{"id":"8b2bf80d7e237c29dc3fe97ff35583a37a3dc195","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-19 20:47:19.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/35/916435/6/20240919211531/NEC-V-ISCSI : SUCCESS in 3h 59m 33s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/35/916435/6/20240919211531/NEC-V-FC : FAILURE in 8h 31m 45s","accounts_in_message":[],"_revision_number":6},{"id":"5327bdc7bb6cea3ddb251fcabb6b46a8ad8a1150","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 01:34:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1146/.","accounts_in_message":[],"_revision_number":6},{"id":"1750a6004e1e3a31a1f8eecbc8c19f2fc885e352","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-20 18:23:39.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/35/916435/6/check/DellEMC_PowerFlex_3.6/6b1694a/DellEMC_PowerFlex_3.6/2119 : SUCCESS in 2h 22m 43s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/35/916435/6/check/DellEMC_PowerFlex_4.x/0a5b018/DellEMC_PowerFlex_4.x/1003 : SUCCESS in 1h 54m 20s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/35/916435/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/ac2a929/DellEMC_PowerFlex_4.x_NVMeTCP/943 : FAILURE in 5h 53m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"f0d8191aac7754e7bed52ad70927cf02e3d53646","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-22 15:29:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/35/916435/6/20240922184303/NEC-ISCSI : SUCCESS in 5h 20m 54s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/35/916435/6/20240922184303/NEC-FC : SUCCESS in 5h 46m 40s","accounts_in_message":[],"_revision_number":6},{"id":"f28824d29d24291cfeac35ad4281c47370b3559b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-24 16:40:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ba97595eae339d9cc230bf3b8ae8b532c9fc9888","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 17:56:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f301069a3b5441c5ad02eae603f104f3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4379222db10449d6aaf6e453cce962fc : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef1bd3f6d89341e29a68aff54c44a167 : SUCCESS in 25m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0a9bc6bb999461bf311fb3bec517fd96f2042c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:09:44.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7da29e1a32bc4b988810da27bc5595e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ca766c3303548cd92f1373306a203fb : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8fd4ce6f38d4c729ee441eb06c9120e : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d428551ec8114d21a7312ed482f0f2d8 : FAILURE in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c4dd706596462d95a67b6bee0507f6 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/66faa05b311a430c96d7a87517e2c48c : SUCCESS in 58m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/877e2f2df94e4e85bba4016c1df2b994 : SUCCESS in 50m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da15673ead77471c84abbf6218193856 : SUCCESS in 1h 42m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0af4019a93654782bb2127dcaa9cca02 : SUCCESS in 40m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/31a1c15b619f4808b324c64738503d2b : FAILURE in 15m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/27015ece2cd84517ac341a8dd51a79f2 : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/967b3eb832b44b05b14beb7e9f686be4 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a40db6baab494aa098f2c381c44674a5 : SUCCESS in 6m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f07900a40b4e46d2bf53952e9df28b21 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7856d2f048eb40a8a2f473cf131fe8f5 : FAILURE in 18m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01bf580f8c7d4c1a8cf27b9a81a88043 : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/138034a5eb9345f19df1196b8e03b8c4 : SUCCESS in 1h 41m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bfb627e630b84e92a4ca698027b32e30 : FAILURE in 2h 23m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13590925d67e4674a0f48d801b5abc45 : SUCCESS in 1h 25m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/78800cddc4ac44878e5b05ff2e0c914f : SUCCESS in 1h 52m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f8d6bbd78d474e7ca2940010e9ea7222 : SUCCESS in 38m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a90371834b0844edb4e6744bd9ac4b8c : SUCCESS in 1h 15m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ffa6b2259f8046aa86d8b3f69f3820dd : SUCCESS in 1h 39m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/647dedea6d1045e8a3f97fe5e3091c77 : SUCCESS in 37m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1ae713ef74de4fa791c27aa8f1325f25 : SUCCESS in 1h 35m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1f688eed49b2433e8a8af81f0f5f516d : SUCCESS in 1h 34m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ec6d9877a7b4257acc5114dd752116e : SUCCESS in 1h 47m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7575f7c9940540f6bd61c887a5419b22 : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":6},{"id":"793468a60217496670b4bfe1a1a347aba70494fc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-24 21:37:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d3c51039eb244cbabe49a8ccaf53d6a0 : FAILURE in 1h 20m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2caeb0ebbf574f76a98af245d02dce96 : SUCCESS in 1h 30m 13s","accounts_in_message":[],"_revision_number":6},{"id":"5028044f32def4ff09017ebf6fd036b0fc982d5d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-21 17:05:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52a8bb3a5454f6e575fc91ed7cff84eeed4990f7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-21 17:11:44.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3819d664c7b3ab198210e0647f1aa240e5c0a652","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 18:34:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abcff66df5874274a749051936fdba77\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71f7fc2e73dc4f75b84b7997a70be301 : FAILURE in 33m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ccbd8b9290bf4653a066f478c7c99e1b : FAILURE in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7c1b6a174152e02682545761f8c7a61501883eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:09:59.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d4f682381944e43983579ceb86a3689\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6ffeefc2ec43638d3fc4700d6f3ad5 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c5e79ba0650434abcc85d1ea72ccb61 : SUCCESS in 13m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c410d412e3d042bfaa463d4bc6b2f654 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd1ba16abfd4b4fb3d1d6812305f8a3 : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/eadc50d9e48a4fc1af784d076aa81547 : SUCCESS in 1h 26m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/862d6d7c2b504787b3081552cb9f3d56 : SUCCESS in 1h 31m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3a5840505be47efa18da505e18395a6 : SUCCESS in 1h 29m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ce62ee1da3848a39c39e4bbc8607131 : SUCCESS in 1h 07m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/798c03693bde48528507a98e3a365015 : SUCCESS in 21m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/29c5ba35ee7e454a9d572d69cdad5fa6 : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0261b605cb604cdd96b1e94d3dbf7c20 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ddbf6513a1304ac09355279def9a222b : SUCCESS in 6m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e0711d0026114e49bec2be5a5a740d6f : SUCCESS in 4m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae8d78c8df834023b2a54b3e22dbbe9b : FAILURE in 35m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/efe6925f98ba4ef795dbc07c55dd67ff : SUCCESS in 8m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1edfb74573864dd182987751e7ec3613 : SUCCESS in 1h 44m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/40f4686a348342a4b234243ee508589f : SUCCESS in 1h 37m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7889e1c183b74b349568aad6af7e594e : SUCCESS in 2h 56m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1bad62552f864e61b95da7c528fa9c7d : SUCCESS in 2h 00m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/46bf348dcd1247a2af268df7efb3fc8a : SUCCESS in 1h 15m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a177b573625a41198e7ee1720c630db1 : SUCCESS in 1h 07m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a922fbd0f69a47d9836869bd8bf60942 : SUCCESS in 2h 00m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cf2d39f7c10d47bdbdaaeb57cacafe31 : SUCCESS in 40m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/48fb9b71ebbf4065887381b68c88f443 : FAILURE in 1h 07m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/438cbbc7a4744184afdf26d135618d34 : SUCCESS in 2h 05m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a14ddcbfc8bb44e089a692e4c4a9844c : SUCCESS in 1h 41m 10s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/336f3d17de6446f08472125258e5f240 : SUCCESS in 1h 22m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e036d0ee79f459fa76baf931d033923 : FAILURE in 18m 41s","accounts_in_message":[],"_revision_number":6},{"id":"9dd339f0d348a40bde7bdc15361bbdb25dc16293","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-21 21:00:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a39a9c75b3cb4a998219df50414c43b3 : FAILURE in 1h 55m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6f1052983f4f49cb8fd1d6954402c95a : SUCCESS in 1h 13m 37s","accounts_in_message":[],"_revision_number":6},{"id":"4e429014bdb7f92153de97a6fbd9493df2b500d9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-24 10:25:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1711cd8bc96ac36a2e9247b71abbb010cba091d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 11:02:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e0601d729664690a6d1b0325e92de83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e8eda53b29c41ed8f0e47703ec8b75b : FAILURE in 33m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba7d53cb4e854ce5843f0114002ccf7b : RETRY_LIMIT Ansible setup timeout in 1m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1aa9f8276eb7fea69957d2d69cc53e7730a5dfa9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-24 11:52:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/07622d1d58e748388b3a4313016ce18b : SUCCESS in 1h 11m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0a0a1726308c4115a68cb03dcbe837a1 : SUCCESS in 1h 19m 02s","accounts_in_message":[],"_revision_number":6},{"id":"3d81001783160238fe7fff8a9e0b0cf5068f482c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 12:52:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c1cc83c2c149e2a6d52d835e2ad078\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d190b765c404224830eda7ca83b4b2c : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49c93a4c042643fc8a8b17369d25df4b : SUCCESS in 9m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c04a0b3576d74c39b89807287abbb548 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42de5030494c49e2b6ff411c9a521f9c : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/f49e4e77a4a540fa99b54b6fb3e0d8d3 : SUCCESS in 47m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4d5b838efdc42829b7c9ccbdb7ec187 : SUCCESS in 1h 01m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6adc9f510e4d41a7af17036fbc379cc5 : SUCCESS in 1h 29m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1af856f544ed4ae7b306a6d9c08ec18f : SUCCESS in 1h 12m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/682c3df523e54251b95b9eeb744a7a8b : SUCCESS in 25m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b1d41e8eb03b4caaa36512bf434c0f47 : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30b11612503d4e129e326ecf6d3fd7b5 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea9d56683b1846c39f4fe1ba5c9009c8 : SUCCESS in 4m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f3dc5c73b46041b19bc7817994d29187 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb2b35a622ff43b6a4d9df56d52a62a8 : FAILURE in 28m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ac9016254db4d36bf089aa83c2d4ea3 : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b228679dabd5445d9a75dfd1988cd38f : SUCCESS in 1h 50m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6a66c078a0204da1a6e7868fd797b677 : FAILURE in 2h 15m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/988d4b3ff86b4f78b45ba115f64f9529 : SUCCESS in 1h 57m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bc08d4dd76384d5d8d66c7f60c316839 : SUCCESS in 2h 04m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/33251238fb954215916721df6f490392 : SUCCESS in 38m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c0768d2a22bb4bfd8a0116a1dcc6b46f : SUCCESS in 1h 12m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ef4fda0c0f5c4c4fb5e51e3e8e060c17 : SUCCESS in 1h 59m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ffd5942b85994d7eb59d3a914e5881d2 : SUCCESS in 43m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c814595995ed4bac8cdd2cd020f56d7f : SUCCESS in 1h 30m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ba33d6c7f8d4484dba01421e30d4eb0d : SUCCESS in 2h 09m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7cdf8066c84741cdbda62ca975d289f7 : SUCCESS in 1h 53m 29s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af3283514ec74c38b4c2b7d463228fbf : SUCCESS in 1h 16m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b945ede3aebe47eeab776ebb4c998bc9 : SUCCESS in 46m 44s","accounts_in_message":[],"_revision_number":6},{"id":"f52d6e34c2c9436c30c777006137ffaa2fbf09f2","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-02-24 12:53:31.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"34d263aa6dd73cdbff1ca642236eebbbb8529a56","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-02-24 12:53:39.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"25ea9c30bc2146ac3a813d9ab5c5599b30699987","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-02-24 12:54:20.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"8b9ed2d6d7c471ab1bb4f1f73536d09c49f9df0e","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-02-24 12:54:27.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"38ac2b0af962cfc123047d5818ff55e2d6d5b143","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-24 12:57:47.000000000","message":"Patch Set 6:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/24_02_2025_12_52_31\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"b7430b7c3ff79bbd2c10c3e1c3dededc5518ed5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"fe73c77737ff23ca77ee3999741f383814df3717","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:50:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"63583025d183e9a7d603d11a34b39aa141d96946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"81f106d1c6b90622094067a04e2732be3139e039","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:06:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"7ab737733009a33374af84fd4d63e4456fc34360","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:21:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a40a5254734e1f9430e4b22eae1445\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6021d2b81e964970b509855aea0182d2 : SUCCESS in 13m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c98d8d8f796d49ce83d6739a5cdbbcbb : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0fce3be29b96496a829202f2f19fc924 : SUCCESS in 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"787cbc4b72b29cd3c32910be89a1a2ba704a1482","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 13:44:57.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544 .","accounts_in_message":[],"_revision_number":7},{"id":"78f7c9958609f7f99f9618a8c5c227c93513ab4f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 15:48:31.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544 .","accounts_in_message":[],"_revision_number":8},{"id":"988c9943237836198b8bc3bb50267d6eb68d23c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 16:26:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2248/.","accounts_in_message":[],"_revision_number":7},{"id":"40f26f08a433130210beaca1c1e1f300329a1c87","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 16:53:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916435 : FAILURE in 38m 34s","accounts_in_message":[],"_revision_number":8},{"id":"27e7bdfa0c859dbc99bd23f4b4e769ce2e206472","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 16:59:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0036fd70a09541c39921e2a26700d6e9 : SUCCESS in 55m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9f3b37f3c4704b2682ced3659cee66a8 : FAILURE in 1h 00m 10s","accounts_in_message":[],"_revision_number":8},{"id":"b49944f14cf9dacaabbe9d014aac889c84ecfa79","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 17:10:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2134/.","accounts_in_message":[],"_revision_number":7},{"id":"29b050d04b413540ba59948637ea3a17cd798c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 17:58:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ff2b8c37aea416dada79620adf4002a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e294de6e07544980b7c64dfc09ae399d : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcc9b4fd1c0d4e9195a9587468368465 : SUCCESS in 13m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20dd9152898d46b3b965278c99cae0c4 : SUCCESS in 14m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49261c1d24db493ca332c13d5cf14d36 : SUCCESS in 16m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/1e3a2e8fc4f14fe485ca133404ca4ad4 : SUCCESS in 1h 01m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e8051a66e7e46a793e7ad33719b135c : SUCCESS in 1h 06m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/25e142b9aef647bf9044ec6f9645a280 : SUCCESS in 1h 41m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1e6908b6d95451880db72e840b6f480 : SUCCESS in 1h 01m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c96fdc435e8a4e289df56d00044918c6 : FAILURE in 26m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c8b3f3dcf6514721b70d858dfb818d77 : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/117eb19c8de040a78f09c3a12e308ef1 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8a6178ede8e94718819e7edc60db34a1 : SUCCESS in 6m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/95ada4454bb54a4898f6a03e4a37d96c : SUCCESS in 8m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fe62a073d44d42fc8852aa981b359c4c : FAILURE in 37m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ffa787abe7042de87d929d1f2986730 : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/348feb14a5274d268ed24e9894fd51ed : SUCCESS in 1h 49m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c4036ec95d954d44aded0697ed574d8f : FAILURE in 1h 51m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/15655e27d15247c18e0da1d235cc1440 : SUCCESS in 2h 05m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/80b20ca3894347b0b9356b2c420a7169 : SUCCESS in 1h 59m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fda7da86bedd4e2795bc0475c1ed2a90 : SUCCESS in 36m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f93fddd2588f4a9abd55460e65efbbb5 : FAILURE in 41m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e605d217c17943eb92c696957aa72cae : SUCCESS in 1h 44m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c7a7a656f7b042ca8524045743ef30ce : SUCCESS in 41m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e34d5cfa9b8f4569a1e46f9bc0d1e5e3 : SUCCESS in 2h 07m 26s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/00f1e3b3bc2a4a089a1ce8911f31ffb9 : SUCCESS in 2h 05m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d32d7d0b7b0d49c98e4a3ff42dba95c1 : FAILURE in 1h 34m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80e356d20a5f493e920d2afb8cd0d602 : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":8},{"id":"a0dcb676deff6b0ff0dcabb311596cb9713d5686","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 09:26:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2266/.","accounts_in_message":[],"_revision_number":8},{"id":"9c5da274170ddbf7b9e1414afbffd76e6810e043","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 09:41:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2152/.","accounts_in_message":[],"_revision_number":8},{"id":"92ded81b72d709236e1a0fccbd5eea60f5ce0457","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 14:37:32.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"561c748fbd2b5aba28b213c8bbe0749f0c700893","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 14:37:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b766fef9c379b0ced454b82122073af60f9bb16a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 15:02:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a202df87e5f4875b47b4daeff6a3ed3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/101e7f9905314c4487f0ec0508290c98 : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ffab4b936eb640f0a5c6598a2398da4f : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1d0af08109ce419ea83762a5bcfcb322 : SUCCESS in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ad4c07c873fd66a3e1e299f369ffbf4057626b06","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-30 16:56:26.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5d82acea82534f49b6e9cdb560f76229 : SUCCESS in 43m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a826fef8ca19490e9ea98e23f6efee65 : FAILURE in 59m 17s","accounts_in_message":[],"_revision_number":8},{"id":"6253d51bdf8d7acf42f0c1b48ae01c2cb98138b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 17:28:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/921b87aa6b5e4ec3acb95db05b1b41bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f31acb583041af8cb83bba1076345d : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/533f108ec6dd4f85ba1817bd7a803fb5 : SUCCESS in 9m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed607ea5f7b64897b342a191509fae70 : SUCCESS in 14m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f2dcb3a9414661a286a06148325c9f : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/08f538e07ef74c0095de063d3ba37cc8 : SUCCESS in 59m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e22d902df7364607be28fd08b86ae2d5 : SUCCESS in 36m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e22c3aad3b2b413bad8194fd04785084 : SUCCESS in 1h 48m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b32650b7ba244611a4d27674b5483733 : SUCCESS in 49m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/26dbeac84fb74a19b4b0b797d228742c : SUCCESS in 19m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f64502532cf1440bbe174df498d0c835 : SUCCESS in 3m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/42f1b70cb87c4748ad5b01d23ed79f3b : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12428a4be3cb435998af02ee828ebe98 : SUCCESS in 5m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a57b27c3c9044357a9fe1d4af22b61f4 : SUCCESS in 9m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be849bc4266f4aa78d5c004ef42e4cf5 : FAILURE in 45m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9a1e48341fd4bcf9bbed907f2dfc15d : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f77a2994ac604bd8b2da0e1504d02e48 : SUCCESS in 1h 55m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/341f640299244a6f89e9a4907aaaad6b : FAILURE in 2h 30m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4da99a84ca664873b13ad8603d1fe915 : SUCCESS in 2h 06m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4095dcba69b04c53bd9978b8e97315a6 : SUCCESS in 2h 39m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9d1249a855ef4e48bbfab1193d522204 : SUCCESS in 34m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/77e8836ad19745978731c0d74909106c : SUCCESS in 34m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b2c3a175f4a64547b5332cc365d75058 : SUCCESS in 1h 37m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a15fbaa18de04c5ebb7f16f9bddbf1ab : SUCCESS in 42m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bc8c536b88b345809ee7b16a8faff9ae : SUCCESS in 2h 01m 00s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b0ec5a59686a414c9ffa3d742dc5f4cd : SUCCESS in 2h 00m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3968ff4c8b44480e80eb62565d20e183 : SUCCESS in 1h 45m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2980112d9ff04dae85239facafc0f064 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":8},{"id":"17da9ecdea826476d17b60052f33e345eb120707","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-30 17:39:58.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_10_2025_17_33_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"b8a81e3d33ce6f7b00f9b6716337a0c619f0dd10","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 17:45:11.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"2da4904c6c18b5055e63c4e42c391af52bb53781","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-10-30 17:45:30.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916434 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"e05e3247cba4bc4d074976e2debd6b45cdaa2631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:08:36.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0aaffcc2144009a4ec027f9aa6e5035f27679d5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:16:45.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4aa5d987ecb407ca31b928fd78afdc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01291a1fda3d4443ad65529229adf01e : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51561c358d5d4bac89babc5d966f101c : SUCCESS in 12m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46c8b01b93aa4609b8b1e96b83287f24 : SUCCESS in 14m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baeadb908a7e400ba57fba4cf9993ca2 : SUCCESS in 13m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/572ecedf51fb49a3ac8269390051d024 : SUCCESS in 47m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0014a590f1cd4181a88926fdedcd8fa9 : SUCCESS in 1h 02m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1206877211a64fb989ba597590fb2122 : SUCCESS in 1h 20m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f0dcbf3cdc945eebad3719c83c946e1 : SUCCESS in 1h 03m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8affb8b5413d43c58ab67bf4b4b2d14b : SUCCESS in 52m 18s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a1f808b3d8d48e3846001e1de8091ed : SUCCESS in 1h 52m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d95a3687719742bcaf37d43f53ef415d : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":8},{"id":"efac2b4c4e2ab9dac068fd2131444b7b03756d8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:16:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e1d7cb646764eb6a72b320d6c499baa495189bb0","revisions":{"38aeb4f924d6caca4fd9344f4f44c68968e1546d":{"kind":"REWORK","_number":1,"created":"2024-04-19 14:51:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/916435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/916435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/916435/1"}}},"commit":{"parents":[{"commit":"609b3a416cd32030fb785d10e71f7e64f148ff5b","subject":"api: Move rate limiting middleware to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/609b3a416cd32030fb785d10e71f7e64f148ff5b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 10:24:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 14:43:54.000000000","tz":60},"subject":"api: Remove v2 limits API controller","message":"api: Remove v2 limits API controller\n\nThe v3 controller was completely overriding this, so there\u0027s no need to\nkeep it around. We delete and (mostly [*]) merge the tests together.\n\n[*] They\u0027re not merged completely because it makes this change appear\nfar more complicated than it actually is. We can tackle whatever\nduplication remains later.\n\nChange-Id: I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38aeb4f924d6caca4fd9344f4f44c68968e1546d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38aeb4f924d6caca4fd9344f4f44c68968e1546d"}]},"branch":"refs/heads/master"},"c89c937bc46e104f67991465563539ee9477c801":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/916435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/916435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/916435/2"}}},"commit":{"parents":[{"commit":"df3612e4fb7311b6496070b53a25ec81c3f066d1","subject":"api: Move rate limiting middleware to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df3612e4fb7311b6496070b53a25ec81c3f066d1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 10:24:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:51:11.000000000","tz":60},"subject":"api: Remove v2 limits API controller","message":"api: Remove v2 limits API controller\n\nThe v3 controller was completely overriding this, so there\u0027s no need to\nkeep it around. We delete and (mostly [*]) merge the tests together.\n\n[*] They\u0027re not merged completely because it makes this change appear\nfar more complicated than it actually is. We can tackle whatever\nduplication remains later.\n\nChange-Id: I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c89c937bc46e104f67991465563539ee9477c801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c89c937bc46e104f67991465563539ee9477c801"}]},"branch":"refs/heads/master"},"2a7c4c4d483a5dae51fa99345e7ea9282a94f34b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/916435/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/916435/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/916435/3"}}},"commit":{"parents":[{"commit":"68282fa5c26480505c185f71871ada5a770164d3","subject":"api: Move rate limiting middleware to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68282fa5c26480505c185f71871ada5a770164d3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 10:24:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"api: Remove v2 limits API controller","message":"api: Remove v2 limits API controller\n\nThe v3 controller was completely overriding this, so there\u0027s no need to\nkeep it around. We delete and (mostly [*]) merge the tests together.\n\n[*] They\u0027re not merged completely because it makes this change appear\nfar more complicated than it actually is. We can tackle whatever\nduplication remains later.\n\nChange-Id: I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a7c4c4d483a5dae51fa99345e7ea9282a94f34b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a7c4c4d483a5dae51fa99345e7ea9282a94f34b"}]},"branch":"refs/heads/master"},"3a5d572c51a1315c5bfe00d206c1dcdf1ab0b803":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/916435/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/916435/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/916435/4"}}},"commit":{"parents":[{"commit":"1a60040c070cb729d763374deb6f6a62e8c0e260","subject":"api: Move rate limiting middleware to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a60040c070cb729d763374deb6f6a62e8c0e260"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 10:24:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 09:58:47.000000000","tz":60},"subject":"api: Remove v2 limits API controller","message":"api: Remove v2 limits API controller\n\nThe v3 controller was completely overriding this, so there\u0027s no need to\nkeep it around. We delete and (mostly [*]) merge the tests together.\n\n[*] They\u0027re not merged completely because it makes this change appear\nfar more complicated than it actually is. We can tackle whatever\nduplication remains later.\n\nChange-Id: I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a5d572c51a1315c5bfe00d206c1dcdf1ab0b803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a5d572c51a1315c5bfe00d206c1dcdf1ab0b803"}]},"branch":"refs/heads/master"},"371bd934a3750f81a10c652b107a363c9d422808":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/916435/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/916435/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/916435/5"}}},"commit":{"parents":[{"commit":"f8e875db6048eef6db73576e4303bd4a4f40c324","subject":"api: Move rate limiting middleware to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e875db6048eef6db73576e4303bd4a4f40c324"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 10:24:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"api: Remove v2 limits API controller","message":"api: Remove v2 limits API controller\n\nThe v3 controller was completely overriding this, so there\u0027s no need to\nkeep it around. We delete and (mostly [*]) merge the tests together.\n\n[*] They\u0027re not merged completely because it makes this change appear\nfar more complicated than it actually is. We can tackle whatever\nduplication remains later.\n\nChange-Id: I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/371bd934a3750f81a10c652b107a363c9d422808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/371bd934a3750f81a10c652b107a363c9d422808"}]},"branch":"refs/heads/master"},"7b5544f49e41daa632de4cd50f4e3b8bf4f538e6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/916435/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/916435/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/916435/6"}}},"commit":{"parents":[{"commit":"e57fc8cdf36522c2e1864cf2ffed7334f2bf2549","subject":"api: Move rate limiting middleware to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e57fc8cdf36522c2e1864cf2ffed7334f2bf2549"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 10:24:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 12:52:28.000000000","tz":60},"subject":"api: Remove v2 limits API controller","message":"api: Remove v2 limits API controller\n\nThe v3 controller was completely overriding this, so there\u0027s no need to\nkeep it around. We delete and (mostly [*]) merge the tests together.\n\n[*] They\u0027re not merged completely because it makes this change appear\nfar more complicated than it actually is. We can tackle whatever\nduplication remains later.\n\nChange-Id: I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b5544f49e41daa632de4cd50f4e3b8bf4f538e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b5544f49e41daa632de4cd50f4e3b8bf4f538e6"}]},"branch":"refs/heads/master"},"11f80c300588f3f763211bec052bf565fec00587":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/916435/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/916435/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/916435/7"}}},"commit":{"parents":[{"commit":"b6bc2b4c06f6f356b2dc865871a41a4251a0ce24","subject":"api: Move rate limiting middleware to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6bc2b4c06f6f356b2dc865871a41a4251a0ce24"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 10:24:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:44:14.000000000","tz":60},"subject":"api: Remove v2 limits API controller","message":"api: Remove v2 limits API controller\n\nThe v3 controller was completely overriding this, so there\u0027s no need to\nkeep it around. We delete and (mostly [*]) merge the tests together.\n\n[*] They\u0027re not merged completely because it makes this change appear\nfar more complicated than it actually is. We can tackle whatever\nduplication remains later.\n\nChange-Id: I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11f80c300588f3f763211bec052bf565fec00587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11f80c300588f3f763211bec052bf565fec00587"}]},"branch":"refs/heads/master"},"e1d7cb646764eb6a72b320d6c499baa495189bb0":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/916435/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/916435/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/916435/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/916435/8"}}},"commit":{"parents":[{"commit":"f6ae4100fca8037e1be46995808849b9fe1cd365","subject":"api: Move rate limiting middleware to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6ae4100fca8037e1be46995808849b9fe1cd365"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-29 10:24:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:53:11.000000000","tz":60},"subject":"api: Remove v2 limits API controller","message":"api: Remove v2 limits API controller\n\nThe v3 controller was completely overriding this, so there\u0027s no need to\nkeep it around. We delete and (mostly [*]) merge the tests together.\n\n[*] They\u0027re not merged completely because it makes this change appear\nfar more complicated than it actually is. We can tackle whatever\nduplication remains later.\n\nChange-Id: I50b04eeb4cc9c30a072734cf2ddd8a9ae6951544\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1d7cb646764eb6a72b320d6c499baa495189bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1d7cb646764eb6a72b320d6c499baa495189bb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
