)]}'
{"id":"openstack%2Fcinder~916436","triplet_id":"openstack%2Fcinder~master~I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-30 20:16:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a","subject":"api: Merge v2, v3 volume metadata controller","status":"MERGED","created":"2024-04-19 14:51:42.000000000","updated":"2025-10-30 20:16:53.000000000","submitted":"2025-10-30 20:16:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916436-openapi","meta_rev_id":"dd84e127f76991819864be2c4e10fc45582b5777","_number":916436,"virtual_id_number":916436,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-30 20:16:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-24 12:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-10-30 14:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-10-30 17:44:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 15:10:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 17:31:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 17:52:52.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-19 19:44:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-04-19 20:36:43.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 21:24:08.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-20 02:41:33.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-20 05:58:57.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-20 07:41:21.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-22 16:55:36.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-23 20:00:52.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-27 08:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-05-01 00:14:08.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-05-02 06:40:03.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 02:15:17.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-16 07:25:50.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-05-20 12:40:55.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-09-18 14:15:11.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 17:26:31.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-18 17:37:08.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-09-18 19:12:25.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-19 02:18:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-21 17:14:13.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-24 13:54:40.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-30 14:37:02.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-10-30 17:45:33.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"17d296e925e5057a62a9f9c656411e3303d6154c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 14:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a18c062c0d76b0fc941d7495750a9fec09e17f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:10:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ca5090974b74b5bb3782d23d27767ea\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47245385e3a74f96b9d06e65892d0ad3 : SUCCESS in 11m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/00d5637316b640d58a7fa6b7cbc3dca4 : SUCCESS in 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3be849f56e18a671c1678f159ef09352383f24b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:31:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6704cc705ba4d5e8dfa4b5c1165fa91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f96295f1f8d1445eb71132c23bdab0fd : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9a92b649a4b434eb5680e4a52d02cc2 : SUCCESS in 10m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aaa18c47f89848209fff841868c2ecaa : SUCCESS in 15m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f91085211aa347ef8839213196f6c431 : FAILURE in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae946e597d548828ab5f72dcf9b69ad : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/b3855f724ac145f1b3b7686a90f47fa7 : SUCCESS in 1h 10m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a9511017238c488f9863e29716bed1cf : SUCCESS in 1h 48m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0141dca31e354e1292d9e045890ca096 : SUCCESS in 58m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17aef5b8fa564bff8abc6e7c0a85321f : SUCCESS in 18m 48s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cb5bcc3fd87e41889f19358d7c7c8ddf : SUCCESS in 12m 46s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/45f83ebde04c47fe979378e6a177d93a : SUCCESS in 6m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9fb11187d2b4fd2a75a3eebf385e9af : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/26b3420c4742457189262164d422bd40 : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e69f2c5c7a8d4a70be1cf7edd675001c : SUCCESS in 9m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/faa12e42884a48dfb0f2404c65ccc9ac : SUCCESS in 48m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fcdd7ba9ab064c4dba36bb3a1d3eb064 : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/30bef75633ce4d21835ec63991100acb : SUCCESS in 2h 04m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/84d8708fb1bf43f9ade25f6dbc78c865 : FAILURE in 2h 32m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec0a7ffdfe5b4a8b8bfcf07603172f46 : SUCCESS in 1h 55m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8fec46994e114a5680dbb3cc2264e092 : SUCCESS in 2h 06m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a932168839bb43f1890bc4159cfab695 : SUCCESS in 28m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fdaa940297354cf099c0a9c431177401 : SUCCESS in 1h 04m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbc7c448679f4a21ac1846d021b4c9cf : SUCCESS in 1h 50m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d942194c33e543b69e2b2fb651391020 : SUCCESS in 54m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0915230485914fb9b0ca2aa61c4e9be3 : SUCCESS in 1h 25m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/32bea3c5cec94b5fa8afd3d0356f1e85 : SUCCESS in 1h 36m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28929892b57840cb991b780039686b5f : SUCCESS in 1h 33m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb860cd5fd6245a1bee2317cf3f9bcc9 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e65f94aca11331d1aba1911a427b8074cc87a9c2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-19 17:52:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/1/upstream-check/cinder-tempest-plugin-ontap-fcp/297708f/ : RETRY_LIMIT in 3m 34s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/fc09a72/ : RETRY_LIMIT in 3m 49s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/1/upstream-check/cinder-tempest-plugin-ontap-nfs/3b505a0/ : RETRY_LIMIT in 3m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/67b701e/ : RETRY_LIMIT in 3m 39s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1c6e176626efb93fd01deb0f9e296a824bd55c2b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-19 19:44:03.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/36/916436/1/check/DellEMC_PowerStore_iSCSI/570e791/DellEMC_PowerStore_iSCSI/1249 : SUCCESS in 1h 09m 15s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/916436/1/check/DellEMC_PowerStore_FC/78790fe/DellEMC_PowerStore_FC/1420 : SUCCESS in 1h 04m 49s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/916436/1/check/DellEMC_PowerStore_NFS/65354f2/DellEMC_PowerStore_NFS/939 : SUCCESS in 1h 14m 57s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/916436/1/check/DellEMC_PowerStore_NVMeTCP/dbd8787/DellEMC_PowerStore_NVMeTCP/800 : SUCCESS in 1h 35m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"43d2e1ee1fdda96d9c25ec5b58a885b9606673c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-19 20:36:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916436 : FAILURE in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e04a20b901a72acb7200432eaaf6ccba63df09d0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-19 21:24:08.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/249cd75cfd794fb9b24666c986492aee\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/eb0bcf2ac11d4bd5b3c02bc954387c90 : FAILURE in 1h 25m 45s","accounts_in_message":[],"_revision_number":1},{"id":"049454201aa9d0042513f9b39f49c4d3f396cf93","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-20 02:41:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/916436/1/check/DellEMC_Unity_FC/a0aec86/DellEMC_Unity_FC/1708 : FAILURE in 1h 41m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/916436/1/check/DellEMC_Unity_iSCSI/1d7fb77/DellEMC_Unity_iSCSI/1738 : SUCCESS in 1h 42m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"99e803efcc7be0f3920e62c31a0170f1e713d963","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-20 05:58:57.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-36-916436-1/FC/ : FAILURE in 12h 5m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-36-916436-1/iSCSI/ : FAILURE in 9h 1m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"41ca71ecf68fe1759e5dd4f5bb7b2a862fae1778","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-20 07:41:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/36/916436/1/ : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b06ee6f4c1138ddcdaa556f803a4cbbb3df55f54","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-22 16:55:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/249/.","accounts_in_message":[],"_revision_number":1},{"id":"b2166c7461591a605c0406054f510b0e04aca010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f58a40564771530eb9b2e876c630c415ce966905","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:18:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df4a8911429a4a82a9be84b93e11f0db\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2212337c49d846c7a4d9eca7443da98d : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea23ddf7bab34bc2a033c9bfc8086146 : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0eee41e091fa7dd50a10e51106b3fd3e2fac7431","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 13:43:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916436 : FAILURE in 12m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6a3f21a2a8cd2afc1ae8aefb4dddb16a4e3ac999","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 15:16:18.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/29f497efc45c4cda82d2b7c01c2feed9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a09b11fe0ba040438d5623df7f4bae5f : FAILURE in 42m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b6883c56ed3a2a4b7d7d1f17a4a15ae4c8ac66f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:22:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a652334970064004bcc6c766c2965960\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2a6d274c284717907c66564df33d6a : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0df50018e6974349a67c485c3f07ceab : SUCCESS in 12m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45b4ad70c2794d71bc50eecd2eede7a7 : SUCCESS in 13m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6fa8a79a14b4874ac327dadf89aa245 : FAILURE in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e8aece1acc4f4295c1dc1e91619249 : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/d4f4e019fb3b47a5b2abf8838de8b65d : SUCCESS in 1h 04m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/115047859ac348f3a48d61cfbd7365b8 : FAILURE in 1h 27m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f53857846af4ebdb444aefe8154a276 : SUCCESS in 54m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/77b6b885ad7f4fa287d74a4b53b07543 : SUCCESS in 17m 43s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dffe0ff6981b40ad81232b1e7e39dac9 : SUCCESS in 14m 35s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/038196aa9ec44483b2cf09f85b1d81ca : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/245f700ed968451ca7d09dd6c77d48f7 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f164a2f768bf4407808cdf384a3feff0 : SUCCESS in 6m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c05b7bf7832e43e4910aed035db8926a : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d27fba38b77c4557a0b1e04263bcd818 : SUCCESS in 33m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2d89ecd9c1b54c0083609ad95f992ea8 : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5da03e52d885407ab617a3f1707151a9 : SUCCESS in 1h 56m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ade9152cb18943718fd1d3ec6ce45e47 : FAILURE in 2h 18m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6add9339e3c7426a861f214fe9204cda : SUCCESS in 1h 54m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/20a69cd769b34faa931cc2c2b37a29df : SUCCESS in 1h 46m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/45e1be5009ac404e9044d1e0355fb192 : SUCCESS in 33m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c2db455d60b243a7a8900c43bef951c0 : SUCCESS in 1h 19m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2cf59e18d42a442b952ccf3a5b47402d : SUCCESS in 1h 43m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4852f79f1fab4aa7967d73e9908c23e0 : FAILURE in 59m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9ecc769fb79e46ffb6397417e764d333 : SUCCESS in 1h 24m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0aa1ab9bf89b40f78e448de3ca0affe0 : SUCCESS in 1h 37m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d034a170a41a458fb891e38516a11b70 : SUCCESS in 1h 55m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21b0c3b3fded433cb456155035c563c3 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5d00dc134be02d842adb98b11cfd5b2343e3ce17","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-23 20:00:52.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/36/916436/2/check/fujitsu-eternus-dx-iscsi/88fb049 : SUCCESS in 3h 34m 33s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/36/916436/2/check/fujitsu-eternus-dx-fc/455ac54 : SUCCESS in 3h 26m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1781e046dff25298737354a0eb4b853c05ec56c3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-23 22:46:16.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/916436/2/check/DellEMC_Unity_FC/10d2db6/DellEMC_Unity_FC/1743 : SUCCESS in 1h 11m 30s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/916436/2/check/DellEMC_Unity_iSCSI/7dededf/DellEMC_Unity_iSCSI/1773 : SUCCESS in 1h 32m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1bd0d44ce4aca592a21af6511f4b12992f75f3f8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-24 01:18:26.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-36-916436-2/FC/ : FAILURE in 9h 15m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-36-916436-2/iSCSI/ : FAILURE in 5h 59m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"9762e9bc86341847f167b06aacc112db25b853b6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-24 13:14:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23b3eef312450b44a6979a1fb7c53f6fbf73a73b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:37:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1734a012043d4d3384bbe58c2144e46d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5721513c74f045c39a77e5507b9f0d86 : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a6e41d4c5abb4ddba9a61992e253a927 : SUCCESS in 21m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4020a25e1545f2b59649a1e2ca87207eb24a3d6d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-24 13:46:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/282/.","accounts_in_message":[],"_revision_number":2},{"id":"2e3d2a4782b8edccb1513434b295e04c1d8999f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 15:49:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5203319e03b045d3a1c366727019c153\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b693259158e4454a91228f50abd3cdc : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d062066661bd4a2f8bf0d1c77216f2c1 : SUCCESS in 10m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/409acb5a4f6a4128bb2773c45f76a18c : SUCCESS in 12m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9be67bcb9ef4530b54f04bbdd7e9a8d : FAILURE in 10m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9391ff4c8f054b6c9eed2ae83f1681ba : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/6e3921bdc6d5495e9b784cbb7ffcc081 : SUCCESS in 49m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/160a7365550b40308b3a280d45a61b7d : FAILURE in 1h 40m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50032b725f51427b9b60e3e0bace494e : SUCCESS in 1h 04m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7aece8dff6174bd9a59a1b4d3c979c3e : SUCCESS in 17m 03s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/882378ff0719438c8c139a6be810c4a8 : SUCCESS in 13m 09s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9586447f84ed4fd0b28578b82d7d253f : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/54302a6413df448d96908e27c0f9e348 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/472c7dc84646437a84b9a638b895b791 : SUCCESS in 6m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/eeb20d7e84344072b3f920a3d18d2f04 : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/237a409c6f364d3597e59ade853e6884 : SUCCESS in 53m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f94ad8c5e3804b77a0fce2288a8ff069 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5f877b9a9bf4acc9b04474748ae27e7 : SUCCESS in 2h 02m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/24a333e374f44fb194a81658556c054d : FAILURE in 2h 26m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/093dad454d0f435897fb1153d195212f : SUCCESS in 1h 40m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/724198b35d17421ca2608d34e2e939ac : SUCCESS in 1h 55m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d8bf94e2671444cab9849b2cad8b7fbb : SUCCESS in 40m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c689edeb36a4c04915bc4d447dd7655 : SUCCESS in 1h 09m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e21b4cc3903e4f15934f29975f56bde6 : SUCCESS in 1h 39m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f2dffc5a2627489e87cb1a5e623d4a73 : SUCCESS in 50m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ecd22550b2c436cbe9386e093dde2b2 : SUCCESS in 1h 27m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dd831514b812420998b397a84c105a01 : SUCCESS in 1h 34m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/74d8abbe438b46acbc9a7c4058b2f479 : SUCCESS in 1h 33m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54f27eb46e1341959cd992283378e27d : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":2},{"id":"820dc95cc15c81e20081bf28a2525539dbbc5e34","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-24 16:35:31.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/93cac8d0a4d34f7681874c7cbc28d756\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f0fc5540fb9d4d5bb9b9f348f189fee5 : FAILURE in 1h 25m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f885a115f55f60c6080ff3c6b898f3333a5b7689","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-26 16:02:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28c871cdaa28cd6337e579e38595f7352f429951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:37:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ae16604fac44b9bb013c98dfd6777c7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f615007b39c7458280505d4bf5e2b506 : SUCCESS in 17m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/357c9d08eceb491f970cce9d2a75c5c6 : SUCCESS in 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da7b95591b41da339e9f08b975c4cc7ffdbe4fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 18:47:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe4a8c3c27a74bb6a8eb7c7c545d4b3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7238dd25b7ef4602924b5b45b432e732 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3eabfee55d2841a6b1f8cf0fe53a5726 : SUCCESS in 10m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6ca277a6ff0421bba7715180d664098 : SUCCESS in 12m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17b3fd0af6664abdb2791c33dc721f53 : FAILURE in 11m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460b7528e08541ad84394afebf80289a : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/f67aa93e24bb4f2899743f3e5f8859db : SUCCESS in 59m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eddec5e80138463ab841eff0602e87a3 : SUCCESS in 1h 27m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e773cdb6cfdd492ead3c528dcb2043b9 : SUCCESS in 54m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a113a08a3b5f476abb72f62b7b2e02d0 : SUCCESS in 18m 29s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/62af82ca8745421181b7f63f5e56f683 : SUCCESS in 11m 55s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/84ca13e2e441445f9a96801dfe4b2756 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b6e49394b0544a3d959c1a25a6b78d60 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4fad27133c1c4cbc9a94527cb5e349f5 : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4aa5c693c2364558a26e1340653b8649 : SUCCESS in 7m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/265b05971b7f49f3b6c13038858e9100 : SUCCESS in 52m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a9536f3a12244328a8a503151369ef1 : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0e3c982e8d784c72b4afb6cabc077d2e : SUCCESS in 2h 10m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/11d90250b446416ca3d325ba13304e2b : SUCCESS in 2h 15m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/88ffefa4a7ad4cab8266b63c7533b22a : FAILURE in 1h 52m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3df74584947c463ab322b5a688deb462 : SUCCESS in 2h 06m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3a1e1797fc8d46c0bd8ec1402673d5f7 : SUCCESS in 31m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aeb4964872d240cbb09638f4d5265936 : SUCCESS in 1h 15m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e3b609d7e2f842f2b602cecd35827d0e : SUCCESS in 1h 34m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/90b6f0b8225e4d6a961e197c804315fe : SUCCESS in 42m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/22075bbb898d4a579ba3a252bd367350 : SUCCESS in 1h 32m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a95e56a1fa734568bd0f7ea0aebccf72 : SUCCESS in 1h 40m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2bb1e2dba6f5424593405bedea677be7 : FAILURE in 1h 49m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1c84e3aa5454ee5a9713a2c6087df56 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2fcc945eb0edf324fe079a5b2b3666fbe911d809","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-27 00:47:34.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/166317ae4a644f88b3dcbd6b58735abf\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/47e856353170452abbab31433474aa3c : FAILURE in 1h 36m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e3158bce0b7a4b61153a282b9aede98549b15bd8","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-27 08:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/36/916436/2/fc : SUCCESS in 0h 41m 10s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/36/916436/2/iscsi : SUCCESS in 0h 54m 31s","accounts_in_message":[],"_revision_number":2},{"id":"cf966daf944f7f45edb6b9ee75812bea8e2a29f1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-05-01 00:14:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/916436-1/HPMSA-iSCSI/ : SUCCESS in 3:05:41\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/916436-1/HPMSA-FC/ : FAILURE in 37:19.47","accounts_in_message":[],"_revision_number":1},{"id":"8acd555bd2d7a07486757381204bb7fdaa4e9bd9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-02 06:40:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/2/20240502105248/NEC-ISCSI : SUCCESS in 4h 18m 53s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/2/20240502105248/NEC-FC : SUCCESS in 4h 47m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b40490728fae4ddded25163f8d2079a90a99b5d0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 02:15:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/36/916436/2/check/DellEMC_PowerFlex_3.6/791019f/DellEMC_PowerFlex_3.6/1245 : SUCCESS in 1h 42m 20s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/36/916436/2/check/DellEMC_PowerFlex_4.5/e32f30b/DellEMC_PowerFlex_4.5/186 : SUCCESS in 2h 06m 06s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/36/916436/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/8dd1318/DellEMC_PowerFlex_4.x_NVMeTCP/122 : FAILURE in 2h 54m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6a9597984ddfcb9b35a704686549424d8ebbb127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"514051eb8b7aaa91a665dad27e9c2fec48689ffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:09:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c687327f277f4aaeb9027c8479915961\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af8b6ac7bf83496c801e85d92e828c6c : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7193823c26c421dbbcbe979431feb29 : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c92f4820d50174a4727133516532676e30011a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:27:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bf0d66128a5456988f85c0cb6255089\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f4d338bd814a52885434131b09e3b0 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/757fa3e59c88415f9bf37e52235f16f3 : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59407a0cb2b9471680ac60da20267452 : SUCCESS in 9m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/029eaad00545444f94d32037d8d48ec6 : FAILURE in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e84e0a3ab3d4a79b4f79d5c40a27da6 : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/8d8c31455f2340a3a5374c84bf321ffc : SUCCESS in 1h 05m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a114c25482a4c019dfa2d6d236e5cde : SUCCESS in 1h 25m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f86a90bd2c446a184a4477726f852fb : SUCCESS in 59m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9948a62ccea4ede84953ba5b34543c7 : SUCCESS in 14m 46s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f249ff15a2e7445bbae24fa0367a466d : SUCCESS in 17m 11s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9364d4d702a7432eab92629ebcc82bc2 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79ea9f1827e7454eab8bc5ff965b0e44 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b66ce956ee0f4de4825c7953c6e1d6a1 : SUCCESS in 8m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/13385aa63e134a8189931555a5321805 : SUCCESS in 9m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/afc1bd28bdb04779b18fbd4f2d357675 : SUCCESS in 42m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c180062b4d0c444a8b020f7feb1e311e : SUCCESS in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c0edf91fbff64d589b2f7431123844a9 : SUCCESS in 2h 04m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e54a2a4c709347f49dd08391f2b88274 : FAILURE in 1h 57m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8a76ec6279fd45319f8ff60f5c22dbc2 : SUCCESS in 1h 51m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/12aca8a8a5634eddaf2bc08cad6de230 : SUCCESS in 1h 37m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2f5abbac2c9d4c6fabc214f6464bab2d : SUCCESS in 30m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/315826c439cb4121a751ac0dcde012ec : SUCCESS in 1h 06m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4aae869a19f84b838ccd9cb696afdf8f : SUCCESS in 1h 28m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/55fb164e5c054c38a4cf79b729f89831 : SUCCESS in 48m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1768b7e6824c4853b05342b9a3c3bf84 : SUCCESS in 1h 15m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fd7e9d349b1a4e7687c267a7cce8f24f : SUCCESS in 1h 30m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9199b26d91645508c22ad50a57f43fe : FAILURE in 1h 34m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84d80a5c59914ad2a00ebe7af1b5dc5e : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":3},{"id":"5549aa0cd279e3979dae56d7854990fd01b8b470","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-13 19:57:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916436 : FAILURE in 49m 45s","accounts_in_message":[],"_revision_number":3},{"id":"b485203ace14644b0473ebf8f2374e54bb9ada45","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-13 20:30:48.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/db12925530774ff79b2f83066e731696\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bb16cd60e54d4cce9b7c31895acdbd24 : FAILURE in 1h 28m 53s","accounts_in_message":[],"_revision_number":3},{"id":"8e7cd178733761d9306ef6287c1885069558610b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-13 21:24:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/365/.","accounts_in_message":[],"_revision_number":3},{"id":"50bf08a832d97a6d6474dbc43af090e9d0576a14","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-14 02:31:07.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/916436/3/check/DellEMC_Unity_FC/9134719/DellEMC_Unity_FC/1846 : SUCCESS in 1h 32m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/916436/3/check/DellEMC_Unity_iSCSI/1c8ecd6/DellEMC_Unity_iSCSI/1874 : SUCCESS in 1h 54m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"341677efca7de5bbb1f180a5e3675717dd49a8fe","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-14 16:00:08.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/36/916436/3/fc : SUCCESS in 0h 38m 49s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/36/916436/3/iscsi : SUCCESS in 0h 45m 10s","accounts_in_message":[],"_revision_number":3},{"id":"ef76694f19775736000a99569a5792562adbf30a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 03:50:01.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/36/916436/3/check/DellEMC_PowerFlex_3.6/773af28/DellEMC_PowerFlex_3.6/1328 : SUCCESS in 1h 50m 51s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/36/916436/3/check/DellEMC_PowerFlex_4.5/67da588/DellEMC_PowerFlex_4.5/253 : SUCCESS in 2h 08m 05s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/36/916436/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/254499d/DellEMC_PowerFlex_4.x_NVMeTCP/200 : FAILURE in 2h 55m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"6980b50610a410933ddd5f0745ea45623081e7c9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 06:19:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/3/20240515062420/NEC-ISCSI : SUCCESS in 4h 13m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/3/20240515062420/NEC-FC : SUCCESS in 8h 54m 49s","accounts_in_message":[],"_revision_number":3},{"id":"f8d8cbc4936f3e9248f647cc957d23294897e993","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-16 07:25:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/3/20240516125620/NEC-V-ISCSI : SUCCESS in 3h 23m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/3/20240516125620/NEC-V-FC : SUCCESS in 3h 29m 28s","accounts_in_message":[],"_revision_number":3},{"id":"103345f7ed8fdd08bbe566b0ac956968312ed409","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9df7ddb3c520d9bc8411541356d7cabf68bbc741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"978c4904e7f719c4bf823754a68bc93e1d5e25ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:28:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780b73ba3f95471c8cc4b0eebf143f8a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/499cd29e98364c32a69c7b2f126aa014 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/311e2e3afe4648c38d84398eb6b95979 : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"72ff149a3895ba2ab65c93d8729929b65f065bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:10:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0238db91f4f2429db7504071e8199f3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737d5b94a5f9442d856eb04254334a52 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5da7b7fd94104bb39e0b06187cfb2b34 : SUCCESS in 11m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/105a26136e4045fcab03db0cfe636924 : SUCCESS in 10m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b10df32488041cabea3335a7500dc68 : FAILURE in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153a65d4926b4a21b055f10063a757ee : SUCCESS in 13m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/e46082f4b411423089265c3992c02dfc : SUCCESS in 1h 18m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49359bc29cf34efeae5d661bf264bf86 : SUCCESS in 1h 42m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcb30e4e253749209679924371c6b36b : SUCCESS in 1h 05m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/623988ff07194fe9a00d6bc8daaaf76e : SUCCESS in 16m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/540f0dfe02264ade84dcc30045c23fc7 : SUCCESS in 5m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/deed6497772946d9a0b4fac034c54c9d : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ceedfc87e99049fba0d82eba7df4a3ed : SUCCESS in 6m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8b173ee7f57749df9a4493a4b7b8f33f : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d72db543a35e41be89bc6b6486d447a9 : FAILURE in 53m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a50254b2fa8543ad83f588bcacbdac60 : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3481ee21b6a9474db35190513bfe5b91 : SUCCESS in 1h 51m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a51952f45ff946c79f7327de157d11a9 : FAILURE in 1h 58m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0331009c8d141b8ac405350c789e787 : SUCCESS in 1h 47m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4f329667470845cbbfe55e2328d1039d : SUCCESS in 1h 38m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2dd8d948ec824c72bfb32cb72adf425b : SUCCESS in 45m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce2cdd11273148cabf160ba584bc3d22 : SUCCESS in 1h 25m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6fbf3bbb5614bb0be1596b2b2cb6eb0 : SUCCESS in 1h 36m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/98135c20d03f4533919d219f237dbbcc : SUCCESS in 47m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8ee877ccb0144e2486c0a64d297cced6 : SUCCESS in 1h 27m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8fc92e77be0c48778d7e7c13a95e2fb3 : SUCCESS in 1h 16m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f78b17f908824814a8f95d2433709ea9 : SUCCESS in 1h 20m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc67a29615eb44d887e490e4672f78e8 : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":5},{"id":"714d5ea50419f21aa91deda294d3c6fc82602629","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-20 12:17:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/5/upstream-check/cinder-tempest-plugin-ontap-fcp/95e42c5/ : RETRY_LIMIT in 4m 54s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/7994a834ce954daf83a26fea46922943 : RETRY_LIMIT in 1m 49s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/5/upstream-check/cinder-tempest-plugin-ontap-nfs/85e2681/ : RETRY_LIMIT in 5m 12s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/582a4fb/ : RETRY_LIMIT in 5m 03s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"cd8018bf905a7cc20407c645c516d0a8336737dc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 12:40:55.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916436 : FAILURE in 23m 14s","accounts_in_message":[],"_revision_number":5},{"id":"52cc072474d57780bfee72f6553871569f86947c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 14:07:08.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6c37ebb51cbf4bf3859a34599a62af90\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/85dfa5f15ab145eaae01cd705f34ab60 : POST_FAILURE in 21m 31s","accounts_in_message":[],"_revision_number":5},{"id":"f1fc6f5d610b0d8671b470b1b8ddba2a08c3c8ab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 14:23:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916436 : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":5},{"id":"285147e0ed8f37b86f06ec88e7cfc4d8c51b1937","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-20 22:41:34.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/36/916436/5/check/DellEMC_PowerStore_iSCSI/31079f6/DellEMC_PowerStore_iSCSI/1346 : SUCCESS in 1h 15m 17s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/916436/5/check/DellEMC_PowerStore_FC/b7cc51f/DellEMC_PowerStore_FC/1529 : SUCCESS in 1h 01m 49s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/916436/5/check/DellEMC_PowerStore_NFS/6f3b1d0/DellEMC_PowerStore_NFS/1040 : SUCCESS in 1h 08m 55s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/916436/5/check/DellEMC_PowerStore_NVMeTCP/c8c38d3/DellEMC_PowerStore_NVMeTCP/894 : SUCCESS in 1h 34m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"b1afae07e40d59416af17217f0740957ba65445f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-21 00:25:49.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-36-916436-4/FC/ : FAILURE in 11h 24m 9s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-36-916436-4/iSCSI/ : SUCCESS in 5h 19m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"790bec1d418c4318aed458c6b96018bd10d1fdce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-21 02:08:25.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/916436/5/check/DellEMC_Unity_FC/c800993/DellEMC_Unity_FC/1906 : FAILURE in 2h 16m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/916436/5/check/DellEMC_Unity_iSCSI/b8ad0d2/DellEMC_Unity_iSCSI/1934 : FAILURE in 2h 35m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"24540b7ee87895faddbc808ec8c3c4e93ecf8da6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-22 03:03:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/5/20240522071445/NEC-ISCSI : SUCCESS in 4h 24m 39s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/5/20240522071445/NEC-FC : SUCCESS in 4h 48m 59s","accounts_in_message":[],"_revision_number":5},{"id":"13214415173afff535f1710c5303870f89aa5c24","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-22 17:24:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/5/20240522224510/NEC-V-FC : SUCCESS in 3h 38m 35s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/916436/5/20240522224510/NEC-V-ISCSI : SUCCESS in 3h 38m 57s","accounts_in_message":[],"_revision_number":5},{"id":"7eec87bfac22931a29752eaf9e85fa392314a066","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-22 22:00:06.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/36/916436/5/fc : SUCCESS in 0h 40m 50s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/36/916436/5/iscsi : SUCCESS in 0h 47m 51s","accounts_in_message":[],"_revision_number":5},{"id":"cfd62a7a1defe301a590e2e9e9142f5a6900c7c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"68b08d3208f23c5f523d2212d1520501be097c87","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:11.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/583/","accounts_in_message":[],"_revision_number":6},{"id":"2bf4154364f057e3b4e05913364bf7527109687b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:13.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916436/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916436/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"6d4d2ff75d44c6d622bb8c32a85233d6ccc07dd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:54:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d0e97e2a2a44bc8d678287aaa66b1f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c5137ac7e7d40c2ac1ef2941a9d4d52 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6eb4743b2d8c43ef933dd5159fefdc52 : SUCCESS in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1a8729f40918e0f638ba92015493309de5854c23","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 17:26:31.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/64c578bc6a1a42a3807660ff3519e444\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/967a50de232146f28684d62f598c0040 : FAILURE in 42m 00s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d2bf2df57c24b92b68d9e997bc48c4f : NODE_FAILURE Node request 200-0000007252 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9bea0e5ae1414a9e90bc984e26e310e0 : FAILURE in 1h 34m 35s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fa63866082a1400ea7de59056b943852 : SUCCESS in 45m 50s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmai... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"10d060a81d57e92b3170d2e731faa7aadd6e34fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 17:31:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b85fc1998745cfba77b081f35ea3e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cdee92608cb4298a8b27806f600d8e6 : SUCCESS in 19m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22634a7b925a4e6f8c448f3671e0458f : SUCCESS in 10m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be09d1face504eef8bdcc5878f003032 : SUCCESS in 10m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a3968b19bb447ab8f77a98e8c4e8e79 : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6233f715565944ed9b0f3a6d735e44f3 : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/37d251d0718f460c9b223308f3c81ea9 : SUCCESS in 57m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf1a9f1e4758422fb5081fc984f68350 : SUCCESS in 1h 17m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34e7826e094f42a6abeb235ec66e0b34 : SUCCESS in 31m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cf0f1ed612124991ab5d73f8dbdba947 : SUCCESS in 16m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9b54572ef8ba464ba2a0a2b2bc139ee1 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0126b76ff2b4117b31dcf21fde3a9e4 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/abca622e62de43fcb67cf5f1467f7348 : SUCCESS in 7m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6b1a38c4eded4a8b9606e0793a06f232 : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/69a0d3f5235f4e4d8b51811c10ff0674 : FAILURE in 34m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/861ed8a740b449608687df8965eb1cc8 : SUCCESS in 16m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/23309052a52b41d2a99e44cfa46625a7 : SUCCESS in 1h 44m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/899b470678f848c9b60bbabfae2a93b1 : SUCCESS in 2h 47m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/04cd9396cdef451cbc246085fcb9ff90 : SUCCESS in 2h 18m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/aedb18208d954087b387da8c83c17ce8 : SUCCESS in 2h 44m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0b45838aab71480c88b6266bfc246276 : SUCCESS in 52m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d6999618aa9840bcb29c6ec6356383eb : SUCCESS in 46m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/562ca14e7af342798924dba0463321fe : FAILURE in 1h 40m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75fa9fbbbcad4f02a56837cf3f624b5c : SUCCESS in 43m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5e571b1ace8d4d45a34d2737421e12bd : SUCCESS in 1h 36m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/87504c0eb10b43a38166bcb84d532611 : SUCCESS in 1h 31m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be3b6ad8593143c59ee6a507d17092a8 : SUCCESS in 1h 01m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c90a5bb0cf184716839ed95189d66580 : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":6},{"id":"5b7d0d9a01d1e399fc1d3138f9203edcaf13b8d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-18 17:37:08.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2024_17_31_54\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"fd169bd8075a1a402d118d5d36b36cea5be6a67a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 18:14:23.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/7ee4210/ : RETRY_LIMIT in 10m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/916436/6/upstream-check/cinder-tempest-plugin-ontap-nfs/75719b3/ : RETRY_LIMIT in 16m 06s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"7d57260b4fc835d524de7b2bdb8a2d8faa38ef7b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-18 19:12:25.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916436 : FAILURE in 27m 10s","accounts_in_message":[],"_revision_number":6},{"id":"5fdc0062ea73e98da1e8fda364da1a795b4b2452","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-18 21:50:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916436 : FAILURE in 58m 14s","accounts_in_message":[],"_revision_number":6},{"id":"89f6b37fd9a0307f7926e8bb9504c49b224d19fa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-19 00:20:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916436 : FAILURE in 1h 17m 15s","accounts_in_message":[],"_revision_number":6},{"id":"771bfe26a00d420f8402ea5142a850ef527b9a25","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-19 02:18:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9b85a1f63d1348ffa5e5addff2669b3e : SUCCESS in 1h 14m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/852c8a8cfaad4a088556514234955a59 : SUCCESS in 1h 26m 06s","accounts_in_message":[],"_revision_number":6},{"id":"d1c24dfc1eacfbfe0638476c1f206b5a497ace01","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-19 04:31:11.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_PowerStore_iSCSI/f4e6427/DellEMC_PowerStore_iSCSI/1751 : SUCCESS in 1h 14m 03s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_PowerStore_FC/bca412d/DellEMC_PowerStore_FC/1949 : SUCCESS in 1h 06m 26s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_PowerStore_NFS/eafe0f3/DellEMC_PowerStore_NFS/1444 : SUCCESS in 1h 11m 22s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_PowerStore_NVMeTCP/dec9d1f/DellEMC_PowerStore_NVMeTCP/1304 : SUCCESS in 1h 58m 35s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_PowerStore_NVMeFC/66e39b4/DellEMC_PowerStore_NVMeFC/268 : SUCCESS in 1h 15m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"f36eaf22e51786a781b70908297f1e82517a745b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 19:58:14.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_Unity_FC/bdfca43/DellEMC_Unity_FC/2626 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_Unity_iSCSI/2ff0a74/DellEMC_Unity_iSCSI/2635 : FAILURE in 2h 33m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"8f7c18d34d33f36e8aada4cc18405e88319b6326","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-19 20:29:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/560/.","accounts_in_message":[],"_revision_number":6},{"id":"35136bd91f4882f3cdd45c81bfdefd3e902420ad","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-20 00:37:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/36/916436/6/20240920054722/NEC-V-ISCSI : SUCCESS in 3h 43m 43s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/36/916436/6/20240920054722/NEC-V-FC : SUCCESS in 3h 49m 50s","accounts_in_message":[],"_revision_number":6},{"id":"18d607c612fbd6e36a6340c0e526645f46c2765a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 04:09:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1148/.","accounts_in_message":[],"_revision_number":6},{"id":"5f15451954079a9aa375d95428d14c63cf4a97dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-21 00:17:38.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_PowerFlex_3.6/9cdf742/DellEMC_PowerFlex_3.6/2120 : SUCCESS in 2h 13m 59s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_PowerFlex_4.x/dc632c8/DellEMC_PowerFlex_4.x/1004 : SUCCESS in 1h 49m 42s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/36/916436/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/5011326/DellEMC_PowerFlex_4.x_NVMeTCP/944 : FAILURE in 5h 53m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"29dd47ff676874fbefd09e6da28b9e4604d648d2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-23 09:55:01.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/36/916436/6/20240923091741/NEC-ISCSI : FAILURE in 9h 04m 57s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/36/916436/6/20240923091741/NEC-FC : FAILURE in 9h 37m 17s","accounts_in_message":[],"_revision_number":6},{"id":"8b48a192aada0373f84d7d7740ffaba633596251","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-11 13:38:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/916436-6/HPMSA-iSCSI/ : FAILURE in 3:27:54\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/916436-6/HPMSA-FC/ : FAILURE in 44:29.85","accounts_in_message":[],"_revision_number":6},{"id":"985be41cedbe85e1ce830afd12c010e45d0c26c8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-21 17:14:13.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a5be5baf0ec68adc31d4b90a43fb0b5b5a3bbdc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"832e9557d1285177ab30cbafa841b71f0b1f7e10","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:50:46.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"c20c2a8bc3ddd73d7f74878578034bcd9507a389","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"3556818ecc4c0f9320d07ea9bb7aabcbb03a68d1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:07:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"6d3656b58e34de24a73ce79d41898be1a7182b59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:32:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8584e8114f42471abe8793725ef02759\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/868d7525329d416a955364428910c4f7 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4aadc2533a7840a0ba832666f59d23fe : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a32643cef52b455ab4302a80f5e976ff : SUCCESS in 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c9928554b61e4bd3d70ffd66aec6844e09a222d9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 13:54:40.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a .","accounts_in_message":[],"_revision_number":7},{"id":"21bf13826c5a0537486f0c4223425ab8dc70ba33","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 15:54:26.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a .","accounts_in_message":[],"_revision_number":8},{"id":"99c61505cf4624ee290ff6f125fe646671fa6935","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 17:29:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2249/.","accounts_in_message":[],"_revision_number":7},{"id":"f6865e95036690a9347d037c19693385fd7f52fe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 17:32:27.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2135/.","accounts_in_message":[],"_revision_number":7},{"id":"36dec21f2a8ff90e1964a874dd6722a0fe18affb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 18:05:32.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/92c3a22c142f4827b4846158a9ab70c6 : SUCCESS in 55m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4384dc2bbf8043819ab0b0b86a405e6c : FAILURE in 1h 06m 01s","accounts_in_message":[],"_revision_number":8},{"id":"f81fff8c3ea2515c3caed9e628bb8dfb6fdcae0b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 18:10:41.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916436 : FAILURE in 1h 12m 49s","accounts_in_message":[],"_revision_number":8},{"id":"78c1a9f0b1d01ca972aeca58bee42ca6dfc50ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 18:45:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9022a1f20a0c47709add62c0e929b5f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa563f9e033948b0aa6cd17964dd47b9 : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df29e4e497b74d32b7a614663a390f3d : SUCCESS in 11m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ac79846bea14b17865d6bb77f6ee2df : SUCCESS in 13m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5b0f5e874e4a0ea22c573fe9e6f35d : SUCCESS in 15m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/8bd8779eebb54a91807f53e1c81c87c0 : SUCCESS in 57m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f9a1fa6dc8e46cfb1717049dee0c16d : SUCCESS in 54m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72588365adea44eaa6fb6d4701280b18 : SUCCESS in 1h 32m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c1a590a880d4d5e9f9094b3ffcff109 : SUCCESS in 1h 07m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8659b7266a8e4451a927a042e5db37e1 : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7dd128757b5a4b679189069960268d01 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa184a74ca3e46a99a5f9e47037b1328 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/239d5b668d3d455582ffc380854cc11b : SUCCESS in 6m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/28703d03d0324a1aa30304df9abe9363 : SUCCESS in 9m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8b5e5e9c88214645bd59834b0e462ade : FAILURE in 36m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb0bccd7ca84412f895df434a513f6e9 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/552fc03c00414cb9bfabcda52cd5bc26 : SUCCESS in 2h 06m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0c14e9774f9443778cd3c878830c8e89 : FAILURE in 2h 28m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/36dba73d97bf4fa6b59bc37d0c57095e : SUCCESS in 1h 33m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d11fcb0aef9b41aeba74998da1502e8b : SUCCESS in 1h 58m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/086a3f1a175f44e3886df6ad0a93e9d8 : SUCCESS in 31m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f5539977dff74f3cafded2d33d599040 : FAILURE in 40m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8fba303f23164957857031c5c086befc : SUCCESS in 1h 34m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/04a1d526cb674859bb9bf8cf6e13049a : SUCCESS in 42m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7a0aacaba1844663828133c381dc6ce7 : SUCCESS in 1h 59m 46s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ed775ee1b9b04b2b801c6d50c152d967 : SUCCESS in 2h 27m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8cffdf451e1b4358bc7955313cc012f7 : FAILURE in 1h 36m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a61ce68045d949599c6a27eb99deafce : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":8},{"id":"7eb9aae4f10705356218a9b42d2e47891c3c8512","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 11:32:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2268/.","accounts_in_message":[],"_revision_number":8},{"id":"6f8a3085dfbe4cb5b4907561153c54fc6b7f148b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 11:42:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2154/.","accounts_in_message":[],"_revision_number":8},{"id":"e42aee9167a4de6e7f54b63c737546728023b398","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 14:37:02.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"be2b2f3cb95c4a1ef9a838b980956fa5787098e8","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 14:37:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"eb817bd754467ecf9ca56abb55e1a4d5875c1dc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 15:12:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f58d9652c36f491a97b4a219bf17c224\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b476fc756ad34c088e7af3397f35d67a : SUCCESS in 24m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5024fd2217f6461d9fecc9138b9944ef : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ef87c72a463c49a7a90526e33d5fda8b : SUCCESS in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d22345294f52eb6c43a58d81baabd395e5cbd401","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-30 15:56:19.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c25fd1aac4af402aa078df69dc2c1016 : SUCCESS in 48m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/af11904ccdf74d3fb1099f42728da57d : FAILURE in 1h 17m 13s","accounts_in_message":[],"_revision_number":8},{"id":"460036d42ecd1502aaa7e58affc1ab0889503816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 17:26:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6214e919027d4c9c89b6f880cb182a3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a271fccd1c164ff2af3506a646fc43c2 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36bbcc08f53f4ad5adbaa55bd45bffd3 : SUCCESS in 11m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77a0a7e683a34afe9856e53f7af48f50 : SUCCESS in 13m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ae887ce31a48bd8f0c584dc213bc1c : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/e0f364195520494a986a09b78787505d : SUCCESS in 1h 04m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc8d09aec67049d086db14ac6414028e : SUCCESS in 1h 04m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7a1bbf6474d4efd8728c46e1ac00aa9 : SUCCESS in 1h 28m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d74c267908bc4dd59cfab956a3a388e0 : SUCCESS in 1h 04m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e07e7b422a8348338652a95a6cb307e9 : TIMED_OUT in 40m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da905605efca448fb23fcfc7220bc4fa : SUCCESS in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81efe26673f649e88061b4c53b29a6f4 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c38afed19cb48e29ae665430a2df030 : SUCCESS in 6m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cf0c4fbcda9c4d16a9afd59462890378 : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0d493177cee9408391afe28ca0b3f89f : FAILURE in 28m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a0014c5ed84d4d8ea3121bc1f13221f7 : SUCCESS in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7585f1dccec84d5988ecebe57982472a : SUCCESS in 2h 17m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd8b0ea79db34e55bb338c96f262233a : FAILURE in 2h 41m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fe0f3384b5d042f8ad69c74029a3eb8f : SUCCESS in 1h 48m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bc6c6e3da3d1440596c7c61fe836f4a7 : SUCCESS in 2h 11m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/79abc9b95a2c41f0a8a77d63a89aa99d : SUCCESS in 31m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d92318982594005995041a9df43bcde : SUCCESS in 1h 14m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/721df75d55ba4e978dc9bea43bcd2b95 : SUCCESS in 1h 22m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6aa26aea6fd04e928cf02dd7790e3d52 : SUCCESS in 31m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4491263c61c24f07adec386e0e5cb6a9 : SUCCESS in 2h 05m 47s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a1261dd26f9343078d54f5e89c19c787 : SUCCESS in 1h 13m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/617b0d0c8ee844f9b60eaa7c6713c3bd : SUCCESS in 1h 53m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f41eb122fdab40b29321643733b4b06f : SUCCESS in 55m 29s","accounts_in_message":[],"_revision_number":8},{"id":"2136da3f5d293e484262354c5914aaa768df2654","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-30 17:33:32.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_10_2025_17_27_01\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"3116f0c5ebdc5f148449f186493eff1d49dc07f6","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 17:44:52.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"64ee4a7c4635a3701533d51cb9e0daf614fd0e04","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-10-30 17:45:33.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916435 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"b47ce21eb19b72b486f12cafd1721b4aa4399b22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:08:40.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8daea712890debc9cd9e007b7e11230ab86bf664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:16:52.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73e4616b55594ffb9d5800e4a53f872b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e8d3b8a9be496cb463664184d79d01 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92201d2c06f147e58726709ddc8628bb : SUCCESS in 11m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3db101aa9b74f6ca37ed938e805b087 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d97acd5e62499da781fd613e46a15f : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/23dda04c88ff436387aab7da7b59bdc2 : SUCCESS in 56m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89ac90ba3f3f4bf48dec9ae8f2470844 : SUCCESS in 59m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e30f1346b0f4e789c1968ae8f16565f : SUCCESS in 1h 30m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f84520ccfd7044efb340c78aa82e0896 : SUCCESS in 1h 03m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/576c3edc7a794f80be6f01d8a2c6cef4 : SUCCESS in 1h 11m 26s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/31173f97c2d3483f882ae4b9d8716d9f : SUCCESS in 1h 56m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c97c4f080d664607b2c79d69d25c8094 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":8},{"id":"dd84e127f76991819864be2c4e10fc45582b5777","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:16:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1cde7143a83c7d00d0c749340ce7dd0f5668b9db","revisions":{"c10bc8dd71dd24e797cbc9c9e6d04cb7324a7580":{"kind":"REWORK","_number":1,"created":"2024-04-19 14:51:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/916436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/916436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/916436/1"}}},"commit":{"parents":[{"commit":"38aeb4f924d6caca4fd9344f4f44c68968e1546d","subject":"api: Remove v2 limits API controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38aeb4f924d6caca4fd9344f4f44c68968e1546d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-18 22:50:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 14:44:53.000000000","tz":60},"subject":"api: Merge v2, v3 volume metadata controller","message":"api: Merge v2, v3 volume metadata controller\n\nThe former is no longer used independently. Merge it with the latter.\n\nChange-Id: I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c10bc8dd71dd24e797cbc9c9e6d04cb7324a7580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c10bc8dd71dd24e797cbc9c9e6d04cb7324a7580"}]},"branch":"refs/heads/master"},"85444e3274a75e0d73e0154511f3df1876402db6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/916436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/916436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/916436/2"}}},"commit":{"parents":[{"commit":"c89c937bc46e104f67991465563539ee9477c801","subject":"api: Remove v2 limits API controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c89c937bc46e104f67991465563539ee9477c801"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-18 22:50:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:51:11.000000000","tz":60},"subject":"api: Merge v2, v3 volume metadata controller","message":"api: Merge v2, v3 volume metadata controller\n\nThe former is no longer used independently. Merge it with the latter.\n\nChange-Id: I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85444e3274a75e0d73e0154511f3df1876402db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85444e3274a75e0d73e0154511f3df1876402db6"}]},"branch":"refs/heads/master"},"65f2f706647fd0379c9acdc149d81ca11c515d45":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/916436/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/916436/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/916436/3"}}},"commit":{"parents":[{"commit":"2a7c4c4d483a5dae51fa99345e7ea9282a94f34b","subject":"api: Remove v2 limits API controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a7c4c4d483a5dae51fa99345e7ea9282a94f34b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-18 22:50:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"api: Merge v2, v3 volume metadata controller","message":"api: Merge v2, v3 volume metadata controller\n\nThe former is no longer used independently. Merge it with the latter.\n\nChange-Id: I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65f2f706647fd0379c9acdc149d81ca11c515d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65f2f706647fd0379c9acdc149d81ca11c515d45"}]},"branch":"refs/heads/master"},"771f9478c21e2d3b0d5de192274b2bd1c5410cf6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/916436/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/916436/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/916436/4"}}},"commit":{"parents":[{"commit":"3a5d572c51a1315c5bfe00d206c1dcdf1ab0b803","subject":"api: Remove v2 limits API controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a5d572c51a1315c5bfe00d206c1dcdf1ab0b803"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-18 22:50:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 09:58:47.000000000","tz":60},"subject":"api: Merge v2, v3 volume metadata controller","message":"api: Merge v2, v3 volume metadata controller\n\nThe former is no longer used independently. Merge it with the latter.\n\nChange-Id: I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/771f9478c21e2d3b0d5de192274b2bd1c5410cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/771f9478c21e2d3b0d5de192274b2bd1c5410cf6"}]},"branch":"refs/heads/master"},"31e0006d87402b5aa4acdbe4462f7cb693b1260e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/916436/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/916436/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/916436/5"}}},"commit":{"parents":[{"commit":"371bd934a3750f81a10c652b107a363c9d422808","subject":"api: Remove v2 limits API controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/371bd934a3750f81a10c652b107a363c9d422808"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-18 22:50:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"api: Merge v2, v3 volume metadata controller","message":"api: Merge v2, v3 volume metadata controller\n\nThe former is no longer used independently. Merge it with the latter.\n\nChange-Id: I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31e0006d87402b5aa4acdbe4462f7cb693b1260e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31e0006d87402b5aa4acdbe4462f7cb693b1260e"}]},"branch":"refs/heads/master"},"000dd4fb1ee16d31964cd99752d2cac27e6ee266":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/916436/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/916436/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/916436/6"}}},"commit":{"parents":[{"commit":"7b5544f49e41daa632de4cd50f4e3b8bf4f538e6","subject":"api: Remove v2 limits API controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b5544f49e41daa632de4cd50f4e3b8bf4f538e6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-18 22:50:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 12:52:28.000000000","tz":60},"subject":"api: Merge v2, v3 volume metadata controller","message":"api: Merge v2, v3 volume metadata controller\n\nThe former is no longer used independently. Merge it with the latter.\n\nChange-Id: I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/000dd4fb1ee16d31964cd99752d2cac27e6ee266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/000dd4fb1ee16d31964cd99752d2cac27e6ee266"}]},"branch":"refs/heads/master"},"bc8ef938e66083d63c044c6890ea6bb3cd76afab":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/916436/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/916436/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/916436/7"}}},"commit":{"parents":[{"commit":"11f80c300588f3f763211bec052bf565fec00587","subject":"api: Remove v2 limits API controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11f80c300588f3f763211bec052bf565fec00587"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-18 22:50:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:44:14.000000000","tz":60},"subject":"api: Merge v2, v3 volume metadata controller","message":"api: Merge v2, v3 volume metadata controller\n\nThe former is no longer used independently. Merge it with the latter.\n\nChange-Id: I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc8ef938e66083d63c044c6890ea6bb3cd76afab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc8ef938e66083d63c044c6890ea6bb3cd76afab"}]},"branch":"refs/heads/master"},"1cde7143a83c7d00d0c749340ce7dd0f5668b9db":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/916436/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/916436/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/916436/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/916436/8"}}},"commit":{"parents":[{"commit":"e1d7cb646764eb6a72b320d6c499baa495189bb0","subject":"api: Remove v2 limits API controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1d7cb646764eb6a72b320d6c499baa495189bb0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-18 22:50:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:53:11.000000000","tz":60},"subject":"api: Merge v2, v3 volume metadata controller","message":"api: Merge v2, v3 volume metadata controller\n\nThe former is no longer used independently. Merge it with the latter.\n\nChange-Id: I4b1299b03dabaf189fefcfecb0d3d70c5bd8db2a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1cde7143a83c7d00d0c749340ce7dd0f5668b9db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1cde7143a83c7d00d0c749340ce7dd0f5668b9db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
