)]}'
{"id":"openstack%2Fcinder~916440","triplet_id":"openstack%2Fcinder~master~If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-31 21:01:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495","subject":"tests: Merge v2, v3 snapshot API tests","status":"MERGED","created":"2024-04-19 14:51:42.000000000","updated":"2025-10-31 21:02:59.000000000","submitted":"2025-10-31 21:01:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916440-openapi","meta_rev_id":"a8c0c8237b68cf3b7ad0a06ed69d1b1e22e46134","_number":916440,"virtual_id_number":916440,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-31 21:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-24 12:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-10-30 14:41:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-10-30 14:41:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 15:13:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 23:25:21.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 23:30:49.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-20 03:06:02.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-04-20 09:18:37.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-20 16:16:13.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-22 23:00:49.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-23 13:55:15.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-24 00:58:07.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-04-24 01:29:02.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 04:31:00.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 09:47:25.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-16 15:54:24.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-05-20 14:30:14.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-23 04:35:00.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-23 19:22:23.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-09-18 14:15:20.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 20:58:21.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-19 07:38:26.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-14 12:30:37.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-10-15 03:14:09.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-02-21 20:48:45.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-24 14:26:45.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-30 14:41:08.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-10-30 17:45:39.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"893779d9ab63408ef4a5d07b52acf9ca294becaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 14:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7cead6c6070d9d19ecbf61ecf6ff115f45841ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:13:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c605cfaef835467abe5b12b26d05f67c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b7aee4b9094363b80d30abf7a3aa5b : FAILURE in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f0ca5b603994b3482adb6e72f9e071f : SUCCESS in 11m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96fdce1c2faa450d9e1966fb3952d739 : SUCCESS in 9m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81d5ebc503a24a0b8293151254793de7 : FAILURE in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c678469b8e14403b84a3251959d4dc1 : SUCCESS in 12m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0d0d72929cea4a6dacba5edc1d2f0799 : SUCCESS in 19m 49s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f1a1524a638427db2f27a4a5f6031b6 : SUCCESS in 12m 39s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc00a7537d63439d9a196a78e869ab07 : SUCCESS in 6m 00s\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/api/v3/test_volumes.py","accounts_in_message":[],"_revision_number":1},{"id":"8d09751015823e20b5bd77be026364b6a7066ee4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:16:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d89056515654bd6a0aa929ec9d223ea\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e40b9dc16bf94b06ad7bd0aec0700df2 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40d95c91b6b4400f901151b051f31ad4 : SUCCESS in 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed0c1dc62ee0068544177515abf1ec4c172fca2a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-19 23:25:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916440 : FAILURE in 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"dc607c94a12739258318fb25e2e9fee52f3b0307","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-19 23:30:49.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8f8ee48691584239bcbff20a92d1e352\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/31b53988e26b4d1e9d18a984db61a830 : FAILURE in 1h 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"23e23a8a30aeb27572c4cf8f406d8d6196f6d9a3","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-20 03:06:02.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/916440/1/fc : SUCCESS in 0h 56m 07s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/916440/1/iscsi : SUCCESS in 0h 59m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cce4f95bd5108533fb8dec7867b887853304ff47","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-20 09:18:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/916440/1/check/DellEMC_Unity_FC/6d2ad17/DellEMC_Unity_FC/1712 : FAILURE in 1h 20m 46s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/1/check/DellEMC_Unity_iSCSI/270abde/DellEMC_Unity_iSCSI/1742 : SUCCESS in 1h 40m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"25f0e36874b01ee708c299fc7c952c82677c30ba","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-20 16:16:13.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-916440-1/FC/ : FAILURE in 17h 48m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-916440-1/iSCSI/ : SUCCESS in 13h 28m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"2538f7027bff9300a1597264210ad998c19349fb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-22 23:00:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/253/.","accounts_in_message":[],"_revision_number":1},{"id":"015c9bbc3f9e316b1b9d57297fc0fe8d9ce5a1fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f4785680eddbb305c0975058c152f145f01d9208","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:14:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32977d772a0d42e2ad871f1e8796f98f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8228fbadb9d64c1c8f08e3c1bb10c877 : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6fce9f9606346fcbfdc925b67845473 : SUCCESS in 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"872f985a09e6c39a0b502495fb40edb5ae48dd5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:48:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e56eb69a7e6450ea29bc73ce17838f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2de8b20e554561b9bff44c5297b1c9 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc01eac9424844478b333d6ec34e6122 : SUCCESS in 13m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a0778493729426996450d1dd2d68e84 : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9f0eb760f344afb91d7be5446eb5762 : FAILURE in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7122116189fb43829904a1466638beac : SUCCESS in 11m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c13e58e35ab4163a347981c5fa25ced : SUCCESS in 16m 17s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/20514a3fe63d46d2bd2e429a7de0c639 : SUCCESS in 12m 14s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/183c3701f8fd4dd6b5bbf876edab4ba3 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b9fe17cc427f97ecdf2f1449ba13ecc929972cc4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-23 13:55:15.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_04_2024_13_49_06\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"60a5a5f1843bd50cc4dfc7054211410b431749ce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 14:09:42.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"84070d1d609c8f578b419e7658a54d05d192a602","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 15:17:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916440 : FAILURE in 13m 08s","accounts_in_message":[],"_revision_number":2},{"id":"cf808741227a53363aad92ce94de69f61dfc9cde","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-24 00:58:07.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_PowerStore_iSCSI/9a8e27d/DellEMC_PowerStore_iSCSI/1263 : SUCCESS in 1h 06m 48s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_PowerStore_FC/2826907/DellEMC_PowerStore_FC/1434 : SUCCESS in 57m 43s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_PowerStore_NFS/d25d24f/DellEMC_PowerStore_NFS/953 : SUCCESS in 1h 07m 07s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_PowerStore_NVMeTCP/e078454/DellEMC_PowerStore_NVMeTCP/814 : SUCCESS in 1h 43m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"846f060f57ffd40dfff0ce18cb1db6d4379c039b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-24 01:29:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/916440/2/check/fujitsu-eternus-dx-iscsi/e8384d4 : ABORTED in 1h 54m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/916440/2/check/fujitsu-eternus-dx-fc/bbcb9a2 : ABORTED in 2h 07m 37s","accounts_in_message":[],"_revision_number":2},{"id":"52794c4645312d85e762b547fab03a325fdf18b9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-24 05:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/916440/2/fc : SUCCESS in 0h 45m 18s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/916440/2/iscsi : SUCCESS in 0h 57m 38s","accounts_in_message":[],"_revision_number":2},{"id":"56f8d2303e58dbde60c5b41a76e67d590b1ab00b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-24 05:12:46.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_Unity_FC/bdcfd20/DellEMC_Unity_FC/1747 : SUCCESS in 1h 20m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_Unity_iSCSI/d9e5b18/DellEMC_Unity_iSCSI/1777 : SUCCESS in 1h 29m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"766a6c0b973b735858df38b703af555c34c414a1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-24 08:09:11.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-916440-2/FC/ : FAILURE in 13h 24m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-916440-2/iSCSI/ : SUCCESS in 9h 50m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"dc543a671835e998c80ce91790a978ca239f6ffd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-24 22:28:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/288/.","accounts_in_message":[],"_revision_number":2},{"id":"c14196e582227e03399f0c41a6a43f8862b58702","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-25 04:31:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/40/916440/2/ : SUCCESS in 1h 22m 27s","accounts_in_message":[],"_revision_number":2},{"id":"fd7af679ddfafa3e08a413ec4af1569bdf0985ea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 09:47:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_PowerFlex_3.6/91253e6/DellEMC_PowerFlex_3.6/1249 : SUCCESS in 1h 53m 57s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_PowerFlex_4.5/8d9fa9b/DellEMC_PowerFlex_4.5/190 : SUCCESS in 1h 51m 12s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/40/916440/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/365a065/DellEMC_PowerFlex_4.x_NVMeTCP/126 : FAILURE in 2h 53m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"802e5772c5360616c2a9e3e27e86cde62f95eec6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"27bc752359ce5b31842c255aa6c0bc2dc6145642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:16:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5395779b3dc42db9a79ef9cbd7a203f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/682adbded70c4d3b8944147d2b9c72f6 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d3858f9c0fc49d3967256640fc46023 : SUCCESS in 21m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3bbca31458b6ee0205fc6a24f86e912f1bc414a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:49:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d9959ab7d214497ba4bde2c6f09dd59\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e12ac730597422ba431b6ed412f5273 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c777c0d562ec481fa37c5e1098e1f9d1 : SUCCESS in 11m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d21c1fadad954c24a8bc3f82d2e9befd : SUCCESS in 10m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f13b173935949718dcc6c7a01882126 : FAILURE in 13m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597bc38ac87c45a9b663c2eb3f486dd6 : SUCCESS in 15m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1d66466c6a70462bb29c1cd3b2faea25 : SUCCESS in 19m 30s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a8a8a8396eb84fefa7f4e084e4a53041 : SUCCESS in 15m 41s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3fdd91372cea43df8471813f1ebb55cf : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":3},{"id":"dbfc4c49bf32cad6b9d98f060a84b50ddc92c74a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-13 22:33:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916440 : FAILURE in 47m 08s","accounts_in_message":[],"_revision_number":3},{"id":"4ba95a553afaddc60b9613946f94feb491302abe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-13 22:36:39.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/968d61ea4b56458ca9e6600837aaa046\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bb927e063e50443ea756d81a3ad12230 : FAILURE in 1h 29m 41s","accounts_in_message":[],"_revision_number":3},{"id":"f25ef02438cd5b7721ae8cb6364a6be7cc95a511","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-14 00:55:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_PowerStore_iSCSI/1565840/DellEMC_PowerStore_iSCSI/1318 : FAILURE in 1h 25m 40s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_PowerStore_FC/e617c62/DellEMC_PowerStore_FC/1488 : SUCCESS in 58m 36s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_PowerStore_NFS/a1e0839/DellEMC_PowerStore_NFS/1008 : SUCCESS in 1h 11m 38s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_PowerStore_NVMeTCP/eab49c5/DellEMC_PowerStore_NVMeTCP/868 : SUCCESS in 1h 46m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"ddaf01e2abde9f1aa6464576e11dc50059437068","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-14 07:12:55.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/372/.","accounts_in_message":[],"_revision_number":3},{"id":"8cda71b626444814361882e5b7428697a5a94aa9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-14 10:02:45.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_Unity_FC/6124cf6/DellEMC_Unity_FC/1850 : FAILURE in 1h 35m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_Unity_iSCSI/bded874/DellEMC_Unity_iSCSI/1878 : FAILURE in 2h 05m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e281d51edfff7139ece8962b4e3aba66e937f192","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-14 23:00:06.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/916440/3/fc : SUCCESS in 0h 42m 12s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/916440/3/iscsi : SUCCESS in 0h 52m 28s","accounts_in_message":[],"_revision_number":3},{"id":"61bd1a29788e8946b3a91e0cfacdb0a108e16399","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 08:13:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_PowerFlex_3.6/5a570bb/DellEMC_PowerFlex_3.6/1332 : SUCCESS in 1h 35m 28s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_PowerFlex_4.5/d616f82/DellEMC_PowerFlex_4.5/257 : FAILURE in 57s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/40/916440/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/e7a2bd2/DellEMC_PowerFlex_4.x_NVMeTCP/204 : FAILURE in 2h 55m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"a40ba28199ebecb8822b5d94929661c4321726c5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-05-16 15:54:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/916440-3/HPMSA-iSCSI/ : SUCCESS in 3:03:46\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/916440-3/HPMSA-FC/ : SUCCESS in 1:40:05","accounts_in_message":[],"_revision_number":3},{"id":"c118b4a06314756b2011f1ad041d8efe7ba8d20f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4e3cb35d63e82f4716e849cbe47781218974293b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"3dc1b96cccac98f5fd49bafb158ffa1e53ae95de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:25:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76819afe47734732b6e1c1658479d73e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d0c2df76bfa4b22a616e6b3b30606fa : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3afe90a3917e4f6c81baf3c6c5bc1d1d : SUCCESS in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"91ee4225cc4c682878b30623b39d2a6e0a3f2485","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:28:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab50c9dca1924c1490659095a7f60cb7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0920b82c39e4f60b8a6cba42ee24a5e : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daf22cd920a546bf9f142c89505a19a5 : SUCCESS in 11m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/592f344be102424ebcee257e14164e26 : SUCCESS in 13m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5e5adb590c84a498feaa1eec10b357b : FAILURE in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea646a901d92423686154204230aaacb : SUCCESS in 12m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d5561eb14b2b4b8eb3e64c82d1ac87cc : SUCCESS in 16m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/616d4283310c4eb5b3e3418b03a60c1a : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":5},{"id":"0143c6902c9e57b33e5723f105a8e42a313693fc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-20 12:03:34.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/5/check/DellEMC_PowerStore_iSCSI/c09a31d/DellEMC_PowerStore_iSCSI/1340 : SUCCESS in 1h 21m 01s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/40/916440/5/check/DellEMC_PowerStore_FC/f9bde70/DellEMC_PowerStore_FC/1523 : SUCCESS in 58m 16s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/40/916440/5/check/DellEMC_PowerStore_NFS/aa539cf/DellEMC_PowerStore_NFS/1034 : SUCCESS in 1h 08m 07s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/40/916440/5/check/DellEMC_PowerStore_NVMeTCP/aa47edd/DellEMC_PowerStore_NVMeTCP/888 : SUCCESS in 1h 34m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"891017d17b15a25073fccc9ea4be23084e1844b2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 14:30:14.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916440 : FAILURE in 22m 46s","accounts_in_message":[],"_revision_number":5},{"id":"3722c327cd81a7bcf9abfffff6b52aa52b096937","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 14:50:08.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2bdf808bb7434943a22a08153d16722b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/712cf816585143bfac70bb0dea40f6d3 : POST_FAILURE in 25m 42s","accounts_in_message":[],"_revision_number":5},{"id":"782eb4f8ef79e467b9b432de6f63e47a5475db98","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 17:20:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916440 : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":5},{"id":"7c0168ac44c3528931cc64edacbd1495b6783477","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-21 01:56:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/433/.","accounts_in_message":[],"_revision_number":5},{"id":"37c091908b8cb5f1427a466166c229cfafd5bd81","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-21 10:55:17.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/916440/5/check/DellEMC_Unity_FC/5db2974/DellEMC_Unity_FC/1910 : FAILURE in 1h 50m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/5/check/DellEMC_Unity_iSCSI/d123d2d/DellEMC_Unity_iSCSI/1938 : FAILURE in 1h 57m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e1b5082f6ba0a855c96fc5222011938f4de41c6c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-23 04:35:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/916440/5/20240523084437/NEC-ISCSI : SUCCESS in 4h 23m 56s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/916440/5/20240523084437/NEC-FC : SUCCESS in 4h 50m 19s","accounts_in_message":[],"_revision_number":5},{"id":"32cba30fd22576f4e8ccc95f329f1a471cc87cb7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-23 19:22:23.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/916440/5/20240523201313/NEC-V-ISCSI : SUCCESS in 7h 19m 41s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/916440/5/20240523201313/NEC-V-FC : FAILURE in 8h 09m 06s","accounts_in_message":[],"_revision_number":5},{"id":"431f21e89d82af52a0ecfd56670d52e953f7e750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"eaf46f1e35027c71dabeddfd0baf0beabac1fa75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:20.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/586/","accounts_in_message":[],"_revision_number":6},{"id":"6fec18f9e3b1c9339069bf7cc7fc80ccaab4e8cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:22.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916440/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916440/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"904136cfee69d0f549ba21d4b834691853479318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 15:36:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a08b0608418e4f4a834169a207f42c7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185addf2e61848fa916fda5e80b16c48 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddb8fcc5419945f29e8b77e2fc62546e : SUCCESS in 14m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53f3a2bffd5a4ac898ae56b02a5a0c62 : SUCCESS in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1def6821ac5c4deb820132520a08b41e : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/194b3f57da764d4bb7b29ca73925ffe1 : SUCCESS in 13m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17266c656f1548beaf31323f311698ad : SUCCESS in 15m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f95fdb2a66414be699d88f1b9393f137 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":6},{"id":"9f6cd8dae32d51cffbed0de100ecbbb6bccdc338","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-18 15:41:59.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2024_15_36_49\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"5735975a4835215f5f1aab6563aefe38c6b452a4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-18 18:51:04.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_iSCSI/b4814a6/DellEMC_PowerStore_iSCSI/1746 : SUCCESS in 1h 52m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_FC/b7464a9/DellEMC_PowerStore_FC/1944 : SUCCESS in 1h 04m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_NFS/cf2e9ce/DellEMC_PowerStore_NFS/1439 : SUCCESS in 1h 12m 11s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_NVMeTCP/a0553d1/DellEMC_PowerStore_NVMeTCP/1299 : SUCCESS in 1h 49m 31s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_NVMeFC/f77f94b/DellEMC_PowerStore_NVMeFC/263 : SUCCESS in 1h 19m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"d0190b16f1b1f14de39073b367a468d475261578","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 20:27:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b2961ac6377473eb8363f4e16007245\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ccb1927728e841b9ab26049480a4b300 : FAILURE in 24m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b3344bde5cf4c95880bc5705ef16f3d : SUCCESS in 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ac657ee08936f835cfbe264b188187c1d16f6ccd","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-18 20:58:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916440 : FAILURE in 19m 22s","accounts_in_message":[],"_revision_number":6},{"id":"71dee383de43be5eaa33b5c7a308e2acf61bac5d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-19 01:18:09.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916440 : FAILURE in 34m 18s","accounts_in_message":[],"_revision_number":6},{"id":"d08bd65ca77c127579ba62e108cc497f451c6284","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-19 02:26:26.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916440 : FAILURE in 24m 48s","accounts_in_message":[],"_revision_number":6},{"id":"3e8ffeba440373754645ebb98560f4114d83dae1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-19 07:38:26.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f75b67b1556942c9b8d5aa73aa67928c : FAILURE in 1h 17m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e402af4a51fb461b88f58d251d58b7ff : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":6},{"id":"6ece0884a91bc9af95e9696a30032e673e1e4030","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-19 17:13:52.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_iSCSI/934bac8/DellEMC_PowerStore_iSCSI/1757 : SUCCESS in 1h 27m 27s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_FC/ca1e0c4/DellEMC_PowerStore_FC/1955 : FAILURE in 2h 53m 06s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_NFS/9d7b779/DellEMC_PowerStore_NFS/1450 : SUCCESS in 1h 07m 39s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_NVMeTCP/e783519/DellEMC_PowerStore_NVMeTCP/1310 : SUCCESS in 1h 58m 33s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerStore_NVMeFC/16c5b9b/DellEMC_PowerStore_NVMeFC/274 : SUCCESS in 1h 11m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"6b090aafe4819a3e608d2f5ab9af316a7addc65c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-19 22:09:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/562/.","accounts_in_message":[],"_revision_number":6},{"id":"5a17d7e2663ba84ba676dd6973ba17f6347f3d91","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 06:46:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1150/.","accounts_in_message":[],"_revision_number":6},{"id":"aac407179a3ec8c68a7ade2f62613707f484a6ef","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-20 07:58:26.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_Unity_FC/0814871/DellEMC_Unity_FC/2630 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_Unity_iSCSI/31b6380/DellEMC_Unity_iSCSI/2639 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"86cc25856886f9c21dba8bdeeb361593af75f3ed","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-21 00:40:03.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/40/916440/6/20240921015804/NEC-V-ISCSI : SUCCESS in 3h 44m 31s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/40/916440/6/20240921015804/NEC-V-FC : FAILURE in 7h 41m 56s","accounts_in_message":[],"_revision_number":6},{"id":"ef26663bc780c3c45c8865687b8424a8e8f95e38","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-21 23:52:07.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerFlex_3.6/3a75ea0/DellEMC_PowerFlex_3.6/2124 : SUCCESS in 2h 21m 01s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerFlex_4.x/ca73e92/DellEMC_PowerFlex_4.x/1008 : SUCCESS in 2h 14m 48s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/40/916440/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/840cc1a/DellEMC_PowerFlex_4.x_NVMeTCP/948 : FAILURE in 5h 52m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"78919ef54d60553f7948fda49cd485ab71e6b2ed","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-25 04:04:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/40/916440/6/20240925045936/NEC-FC : SUCCESS in 5h 42m 39s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/40/916440/6/20240925045936/NEC-ISCSI : SUCCESS in 8h 05m 09s","accounts_in_message":[],"_revision_number":6},{"id":"f38cf1951e9c0b981920fb997d2ee80bbbfc0a42","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-09 00:45:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/916440-6/HPMSA-iSCSI/ : SUCCESS in 3:18:17\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/916440-6/HPMSA-FC/ : FAILURE in 14:59:22","accounts_in_message":[],"_revision_number":6},{"id":"d573f089ed0dad59ea9bc5033b4f17961127c0d7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-10-14 12:30:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/916440-6/Lenovo-iSCSI/ : SUCCESS in 2:23:24\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/916440-6/Lenovo-FC/ : FAILURE in 6:19:02","accounts_in_message":[],"_revision_number":6},{"id":"6b91bf3586105b628fd08c3c2bdfbea77f7fc59e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-15 03:14:09.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"a10ed55369e6877119e9eafb8e4b2e31e96e25a2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-21 20:48:45.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c54de94669751a15a1a719ca9c48b4a16c737232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1aed59964a86ac29257ea8c02cb25d5bb8327c3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"88712e9c0442961f748f1e577eed637eb717de50","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 12:53:48.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"fa1e139a8710090bbe23fee6f16781db9e3d67d0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:54:06.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"97e92f729d77a2e504ca7cb74062ca43d5ffbd9b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 12:54:08.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"32cc58144d9333586442b7ffef4435974cb71938","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:10:19.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"193cd449bc5c37dc1b526aa14496034fe18a90eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:50:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77a1eb43f0384ff39f58448f74bcd17f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a431154c64a642a8aca9cd1c40cfd640 : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d4e5402635d4f93b92c40dcefeae9ce : SUCCESS in 12m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f4f699b93f54e1cab6358ff442835ad : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977bd84ce6b14ed09c18c01b52fab8be : SUCCESS in 13m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74790deaa1894a88965ccecd0e3750cd : SUCCESS in 53m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/77ab27c0bf954506b0863e708f68475e : FAILURE in 26m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ba8a524241ab410589f03d5b9f171f65 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":8},{"id":"193dfcda659c3eb4e96e881912598fc36ba158df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:50:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726162497aad40daab0c00122520c58a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/250ff6e73d3b436793f3d7938f2e158d : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7849d67e9c824b28a4906483dfc08ef3 : SUCCESS in 16m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/534ef973d7b54f8dbb14b8f671ac16ba : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"08e541418f9d6c6b5b908f248e8f479268e0ec9f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-24 14:02:50.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_10_2025_13_56_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"021434b74d7ea8c29d942fda4960614151da580c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 14:26:45.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495 .","accounts_in_message":[],"_revision_number":7},{"id":"6b44c379b0f34c67e81c658606f378c353c1d5a4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 15:55:44.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495 .","accounts_in_message":[],"_revision_number":8},{"id":"42a5656adcacfec692cd023e16de7622adc47af5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 21:25:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2139/.","accounts_in_message":[],"_revision_number":7},{"id":"6c383bd978aec84859b839892c444def704004a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 21:34:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2253/.","accounts_in_message":[],"_revision_number":7},{"id":"86f97ec373e56c22f6b9906c8c3d577fea049cfa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 21:55:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3e57486686e04a5a997bdfd899134348 : SUCCESS in 49m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/212dd82a3733419b8dc5de32d81b97e2 : FAILURE in 1h 05m 59s","accounts_in_message":[],"_revision_number":8},{"id":"18c5e4fdde5875c7e4c50013150e64b7e9b12e63","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 22:02:23.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916440 : FAILURE in 1h 09m 21s","accounts_in_message":[],"_revision_number":8},{"id":"0297b4fd50cefa09906cfddd42ddcc5610336bb0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 12:53:59.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2270/.","accounts_in_message":[],"_revision_number":8},{"id":"6605ac13a172e3ffb3604098b005efc77b2e9ce5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 13:03:13.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2156/.","accounts_in_message":[],"_revision_number":8},{"id":"cfd6984c7cb5e97a1faffe0f337e83b23a5bf473","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 14:41:08.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d98d7a5ec1f95023c602991cfe2901b2542a5575","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-10-30 17:45:39.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916439 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"2d1553cc1442401f8e951a4fb4ed8212b3231a65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:08:41.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"1b058918e97fb04029441b292bea9b0a7e7e0e1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:19:28.000000000","message":"Patch Set 8: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916439 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"be7f504cd13bb36b3ab0508edf63b40516d47ecd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 10:19:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"69a4acb16e59efd69342e836b6372b4e7b0bf2f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 10:50:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40236db375eb4eb198ee5d93f4598ee9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a860fbf38069482f94af423bf0ac8f0a : SUCCESS in 8m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a74b1ba67ad942d5829f36c204fc0a98 : SUCCESS in 8m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/635eca241d8f41ec979f42960ff5d58d : SUCCESS in 15m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e99bd863c314154bd1d79dce983dad0 : SUCCESS in 15m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eef8e4fb1be14105a816d2a8e0baba1b : SUCCESS in 29m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2181142fbbdd414a9601296e184762de : SUCCESS in 15m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d2a85ec5d2fe488f8ac4e2fb5b6c37b2 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":8},{"id":"529bf16aad46277f1abc60793d7ff23dadf4f128","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 11:19:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9eebffbda0b47cd99b8bfd96345ca35\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62a463dbb6fc4656a4f62a6ec343faef : SUCCESS in 23m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c6278840ec34f46a663554ca94dd656 : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/785e051a28a746c5b50f2fd7ecfc220a : SUCCESS in 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"39f2f4d9e64d1986ccdab8884786493bc8993bda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 13:06:00.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5344770bfcc1f0118885e45e3df1cb400bf98e32","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-10-31 15:14:22.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916439 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"add7cf2b7957ed284e136356a7ad237cb66f9fed","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-10-31 15:14:49.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916439 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"d69a8533f2bbfcdae3ef4ca70f0b16817c72c9b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 15:14:50.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb2766d69f53452c8ec06436ab4aa33b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d419590d1f444249a636865e2b6597a : SUCCESS in 8m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bab5657f5f24734a59fe9923658fcab : SUCCESS in 11m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3201f253aa54310b329778b1527fe27 : TIMED_OUT in 40m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522f1feb37a348a5b9625f0306c84b49 : SUCCESS in 17m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ebb2d0451b7443e9874d045d1228afb3 : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":8},{"id":"3f0da28a030f4378a4dc5c20fa88cd71c51e6815","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-31 15:16:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c1cb245104b71ae57bcef24c8a10c5d0091282bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:34:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab6fbb1ac0c488bbe8e00702bf28a35\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93dd1a442b814019b69f9598cc11717b : FAILURE in 14m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/944426033c0a4a8b90e9579c00c61898 : SUCCESS in 19m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59493d92cef042cf85ecac39c6cb9c61 : SUCCESS in 30m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"05c2455b8b92cb39beb811736ed2fe5ac8207223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:39:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/198c0ef0fe9d45a3b0f8affe1869faf6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d6909ca9d274c2591af683da10bde80 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68cc9a06d9d14f5d90b30de1412a9c01 : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7801c0621561458f8315c1cd2865b218 : SUCCESS in 14m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac38c50aa1443639268d55c032d129a : SUCCESS in 14m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f4fd917050704a2caa7981ed877eedd6 : SUCCESS in 1h 16m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7299ce4568a64e15846e7366b1a4dc97 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e53749b5af8442796fd44b6c7993f2a : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":8},{"id":"9ab00e0076090b26344e53c091eb5cea543f859a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:40:08.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"52c0710f1a9f86033cbdc9869ccca05f8ed8bffd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-31 16:45:58.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_10_2025_16_39_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"baf2e53bef6220c9108e2d2b23523e2d5202ee6f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-31 18:46:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6d3ffe04aca94139be0750d5dbea2232 : SUCCESS in 1h 11m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f07db55b8a714c98a61830b4d171c563 : FAILURE in 1h 15m 48s","accounts_in_message":[],"_revision_number":8},{"id":"cb0fed236e92e7e0318a382d6e231df81730b617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:01:11.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c66eed420b4e49eba975b8ba2d4d82be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22904dd1c53478bb80c34ba3cfc8be3 : SUCCESS in 7m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06bafb804918419b8ab633c643dd7f89 : SUCCESS in 12m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4001a5f13604409ea4dee0e95f726661 : SUCCESS in 16m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a068681e40194727a3c77b2233b8612c : SUCCESS in 16m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e787c3fff99641ba89a23d5f99234299 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":8},{"id":"4cfb4986b41ed3b611eceb42e07ab47893f4d0ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:01:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"a8c0c8237b68cf3b7ad0a06ed69d1b1e22e46134","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:02:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b265de80caf44a08b750b4247d375eff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acefbff37ccb40b698f30086f6f4fb5f : SUCCESS in 47s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"ae1d24afc41c079742c7556bcd3cd6fe9dcdf407","revisions":{"4e4e27cc40a49c5796e1297f69f66d76d7a6e448":{"kind":"REWORK","_number":1,"created":"2024-04-19 14:51:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/916440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/916440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/916440/1"}}},"commit":{"parents":[{"commit":"a272223cd329ea6bb5800459307043a9c0806e4d","subject":"api: Merge v2, v3 volume controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a272223cd329ea6bb5800459307043a9c0806e4d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 11:42:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 14:46:57.000000000","tz":60},"subject":"tests: Merge v2, v3 snapshot API tests","message":"tests: Merge v2, v3 snapshot API tests\n\nWe no longer support the v2 API. Given the v3 controller is the v2\ncontroller but with the addition of microversions and some extra routes\n*and* the v2 tests obviously don\u0027t set microversions or test any of\nthose additional routes, we can simply point the v2 controller tests at\nv3 controller and things will just work (TM). We do just this, moving\nthe test to the v3 test file in the process.\n\nChange-Id: If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e4e27cc40a49c5796e1297f69f66d76d7a6e448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e4e27cc40a49c5796e1297f69f66d76d7a6e448"}]},"branch":"refs/heads/master"},"f44f23727d40cf51b8696913dfd63fec91c0cfbf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/916440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/916440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/916440/2"}}},"commit":{"parents":[{"commit":"cf518fda781903448d6cfb73eabf56387942f4a5","subject":"api: Merge v2, v3 volume controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf518fda781903448d6cfb73eabf56387942f4a5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 11:42:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:51:11.000000000","tz":60},"subject":"tests: Merge v2, v3 snapshot API tests","message":"tests: Merge v2, v3 snapshot API tests\n\nWe no longer support the v2 API. Given the v3 controller is the v2\ncontroller but with the addition of microversions and some extra routes\n*and* the v2 tests obviously don\u0027t set microversions or test any of\nthose additional routes, we can simply point the v2 controller tests at\nv3 controller and things will just work (TM). We do just this, moving\nthe test to the v3 test file in the process.\n\nChange-Id: If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f44f23727d40cf51b8696913dfd63fec91c0cfbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f44f23727d40cf51b8696913dfd63fec91c0cfbf"}]},"branch":"refs/heads/master"},"f49b4cd224382b64a291d33b3263a315c89b29d1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/916440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/916440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/916440/3"}}},"commit":{"parents":[{"commit":"b415ed612541584e4d50afb6d4dcd0b40a003b7d","subject":"api: Merge v2, v3 volume controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b415ed612541584e4d50afb6d4dcd0b40a003b7d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 11:42:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"tests: Merge v2, v3 snapshot API tests","message":"tests: Merge v2, v3 snapshot API tests\n\nWe no longer support the v2 API. Given the v3 controller is the v2\ncontroller but with the addition of microversions and some extra routes\n*and* the v2 tests obviously don\u0027t set microversions or test any of\nthose additional routes, we can simply point the v2 controller tests at\nv3 controller and things will just work (TM). We do just this, moving\nthe test to the v3 test file in the process.\n\nChange-Id: If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f49b4cd224382b64a291d33b3263a315c89b29d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f49b4cd224382b64a291d33b3263a315c89b29d1"}]},"branch":"refs/heads/master"},"b516a515850328f2cb1f406a9eb4a0b1d26945b7":{"kind":"REWORK","_number":4,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/916440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/916440/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/916440/4"}}},"commit":{"parents":[{"commit":"d2a83e046211d5d93765558e7d0fb64bed5d8b2a","subject":"api: Merge v2, v3 volume controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d2a83e046211d5d93765558e7d0fb64bed5d8b2a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 11:42:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:00:16.000000000","tz":60},"subject":"tests: Merge v2, v3 snapshot API tests","message":"tests: Merge v2, v3 snapshot API tests\n\nWe no longer support the v2 API. Given the v3 controller is the v2\ncontroller but with the addition of microversions and some extra routes\n*and* the v2 tests obviously don\u0027t set microversions or test any of\nthose additional routes, we can simply point the v2 controller tests at\nv3 controller and things will just work (TM). We do just this, moving\nthe test to the v3 test file in the process.\n\nChange-Id: If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b516a515850328f2cb1f406a9eb4a0b1d26945b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b516a515850328f2cb1f406a9eb4a0b1d26945b7"}]},"branch":"refs/heads/master"},"657d3b59664416182f4a3ccc62a0027a58752f70":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/916440/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/916440/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/916440/5"}}},"commit":{"parents":[{"commit":"2a213fb53c1e15b42653bf78691f8f04934d1e5b","subject":"api: Merge v2, v3 volume controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a213fb53c1e15b42653bf78691f8f04934d1e5b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 11:42:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"tests: Merge v2, v3 snapshot API tests","message":"tests: Merge v2, v3 snapshot API tests\n\nWe no longer support the v2 API. Given the v3 controller is the v2\ncontroller but with the addition of microversions and some extra routes\n*and* the v2 tests obviously don\u0027t set microversions or test any of\nthose additional routes, we can simply point the v2 controller tests at\nv3 controller and things will just work (TM). We do just this, moving\nthe test to the v3 test file in the process.\n\nChange-Id: If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/657d3b59664416182f4a3ccc62a0027a58752f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/657d3b59664416182f4a3ccc62a0027a58752f70"}]},"branch":"refs/heads/master"},"24f8a37052bc2c1f3f524aad31532285a5607d69":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/916440/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/916440/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/916440/6"}}},"commit":{"parents":[{"commit":"f47a41930859fe805f5afbccc4eda5dfe84a1002","subject":"api: Merge v2, v3 volume controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f47a41930859fe805f5afbccc4eda5dfe84a1002"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 11:42:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 12:57:11.000000000","tz":60},"subject":"tests: Merge v2, v3 snapshot API tests","message":"tests: Merge v2, v3 snapshot API tests\n\nWe no longer support the v2 API. Given the v3 controller is the v2\ncontroller but with the addition of microversions and some extra routes\n*and* the v2 tests obviously don\u0027t set microversions or test any of\nthose additional routes, we can simply point the v2 controller tests at\nv3 controller and things will just work (TM). We do just this, moving\nthe test to the v3 test file in the process.\n\nChange-Id: If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24f8a37052bc2c1f3f524aad31532285a5607d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24f8a37052bc2c1f3f524aad31532285a5607d69"}]},"branch":"refs/heads/master"},"f162dc5b6eb4efa4499896ffa8939351ea1363a9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/916440/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/916440/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/916440/7"}}},"commit":{"parents":[{"commit":"373b8fdae2645abbc3912f2f70a535640d552db7","subject":"api: Merge v2, v3 volume controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/373b8fdae2645abbc3912f2f70a535640d552db7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 11:42:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:47:28.000000000","tz":60},"subject":"tests: Merge v2, v3 snapshot API tests","message":"tests: Merge v2, v3 snapshot API tests\n\nWe no longer support the v2 API. Given the v3 controller is the v2\ncontroller but with the addition of microversions and some extra routes\n*and* the v2 tests obviously don\u0027t set microversions or test any of\nthose additional routes, we can simply point the v2 controller tests at\nv3 controller and things will just work (TM). We do just this, moving\nthe test to the v3 test file in the process.\n\nChange-Id: If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f162dc5b6eb4efa4499896ffa8939351ea1363a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f162dc5b6eb4efa4499896ffa8939351ea1363a9"}]},"branch":"refs/heads/master"},"ae1d24afc41c079742c7556bcd3cd6fe9dcdf407":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/916440/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/916440/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/916440/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/916440/8"}}},"commit":{"parents":[{"commit":"4ac0c90bac99ea72d1ec987b29f8503f210d2e83","subject":"api: Merge v2, v3 volume controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ac0c90bac99ea72d1ec987b29f8503f210d2e83"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 11:42:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:53:11.000000000","tz":60},"subject":"tests: Merge v2, v3 snapshot API tests","message":"tests: Merge v2, v3 snapshot API tests\n\nWe no longer support the v2 API. Given the v3 controller is the v2\ncontroller but with the addition of microversions and some extra routes\n*and* the v2 tests obviously don\u0027t set microversions or test any of\nthose additional routes, we can simply point the v2 controller tests at\nv3 controller and things will just work (TM). We do just this, moving\nthe test to the v3 test file in the process.\n\nChange-Id: If90d2029fec5a5f5fce95a2ff7cbad2bdc71e495\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae1d24afc41c079742c7556bcd3cd6fe9dcdf407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae1d24afc41c079742c7556bcd3cd6fe9dcdf407"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
