)]}'
{"id":"openstack%2Fcinder~916464","triplet_id":"openstack%2Fcinder~master~I4ab65288641638fc489bf2b1d293945c099522ec","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-11-07 02:28:13.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-07 04:23:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ab65288641638fc489bf2b1d293945c099522ec","subject":"api: Remove unnecessary router split","status":"MERGED","created":"2024-04-19 14:51:42.000000000","updated":"2025-11-07 13:32:16.000000000","submitted":"2025-11-07 04:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916464-openapi","meta_rev_id":"86f2e59d683383d100addf8f413b42e0ac93b8cb","_number":916464,"virtual_id_number":916464,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-07 04:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-11-07 02:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-10-31 15:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-11-07 02:28:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 15:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 19:53:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-20 03:01:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-20 03:51:40.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-20 07:00:09.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-04-20 19:22:17.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-21 03:39:41.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-22 13:23:03.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-22 20:36:47.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-04-23 04:55:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-24 05:16:13.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 13:49:40.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-04-26 18:49:30.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-26 19:08:36.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-26 22:07:57.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-10 20:15:03.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-18 14:15:42.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 19:33:40.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-18 23:25:51.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-19 13:06:59.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-12 08:55:55.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-10-24 14:55:16.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-30 15:10:01.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2025-10-31 15:19:04.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-10-31 17:28:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"d17eb40e697854d7cfa578deafbbec953d8cfe0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 14:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c262e4782b9006cdd24cf3527631d91980eccc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:14:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b53ca0e8a454993b3d17588aa7a9b35\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c5383345f1c4f45abe4c29d4081a190 : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f6c37c9854854c189aa7ed44bc8fe502 : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45a54617165df9293dcfb1f3513fecfe009d8c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 17:27:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a5181d5b93c271378bfcfefec7b6a87004c9b6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:38:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1200d64faa2e44a08916c86e933b7862\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66e8fa96330646d7b38167e5dfb404f1 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0bc8b749b9df4d8887b537427103d657 : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ae31102cb601b4a086b2c5a7310a7d93b852bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 19:53:54.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/023edb54e52741eda38e676f13ec8992\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e4d45c14f1b45cfbe042154d7650cc3 : FAILURE in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/210621d2521540fa9f5d5d0719b5a9c3 : SUCCESS in 9m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e804729729a448ba18cc22d1b8f3b47 : SUCCESS in 10m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76d0d9d435204bc8b308c8f3c37271fa : FAILURE in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31832949570a4cd6a5ed0b7d5851377d : SUCCESS in 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/307c5d4d98aa43d39c8c6ac8de2e8600 : SUCCESS in 45m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad4a52c642b14984847e53681f49a505 : SUCCESS in 1h 25m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7a2612bd6934a17aee7866cb3e9d028 : SUCCESS in 1h 00m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e1d55439ddf344e5b5198eb9c1d11465 : SUCCESS in 13m 56s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c0309f5ad0644798b397c04fd3038002 : SUCCESS in 10m 05s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f730d2cac2843a0833d3fcd35b0123e : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f99a580c19ec4f4997b6210a7cc94d18 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0fa32616e1d64c09a0ade56f44fecd6d : FAILURE in 6m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3bc2b0f7b09488dbdda7c19132045b6 : FAILURE in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e91379fb837f4f39b800b671884bd02a : SUCCESS in 30m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fbd687e48d4b4b15b4e7318f0e3c8829 : SUCCESS in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/634df331ef354fabb2ea98853fe3769a : SUCCESS in 1h 28m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/042f7e52d9c54626a70300f8d82dd2dd : FAILURE in 2h 18m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ddffcf3d90da4653ac0784e261915a79 : SUCCESS in 2h 09m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f3b543637eb44589912191dc7f7d65fa : SUCCESS in 1h 59m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bb001bf5ffce4ff79c77f98c18050146 : SUCCESS in 39m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/681bdc1ec0714453b7cb7e5e963e3a22 : SUCCESS in 1h 14m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/60deb78387e24fb49e1c558ce084bd63 : SUCCESS in 1h 38m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9f4ce76f5bae47939227b14303b8fd3a : SUCCESS in 43m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2d404f9941c34632b0b95ebe84b8993c : SUCCESS in 1h 24m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/055d361df8374cb288eb8a5e47a41212 : SUCCESS in 1h 21m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a646a01f86144dabb8f26ade225896e2 : SUCCESS in 1h 39m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87b565da64d74d08ac4a36c6edd7983b : SUCCESS in 1h 07m 16s\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/api/v3/test_snapshot_metadata.py\n  Comments left for invalid file cinder/tests/unit/api/v3/test_volumes.py","accounts_in_message":[],"_revision_number":2},{"id":"91505cc1ba1551c7c9511643bb38d0dc13036616","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-20 03:01:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916464 : FAILURE in 43m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f29c81083652bfba84453d5d5ddadde67c70e34d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-20 03:51:40.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/639664e6678b4337b45b1af901412dc6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c6a3d7ac375f42728447e26c8d477115 : FAILURE in 1h 24m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3865ec33a3f6bf9db3f1b1dff1cca3564c306003","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-20 07:00:09.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/916464/2/fc : SUCCESS in 0h 36m 11s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/916464/2/iscsi : SUCCESS in 0h 58m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4a9a0182a7be86f6279c3dfa8a4cbf511bf7bc0a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-20 19:22:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/64/916464/2/check/DellEMC_Unity_FC/d63cabf/DellEMC_Unity_FC/1717 : FAILURE in 1h 32m 49s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/2/check/DellEMC_Unity_iSCSI/0e09cb2/DellEMC_Unity_iSCSI/1747 : FAILURE in 2h 06m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"45b90c976ce995e4971a8f0a12ce099c3eab42e2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-21 03:39:41.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-64-916464-1/FC/ : FAILURE in 25h 36m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-64-916464-1/iSCSI/ : FAILURE in 21h 17m 48s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"6f7e4c0612a1f53d193baa126885f1b05b7f9505","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-21 16:41:14.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-64-916464-2/FC/ : FAILURE in 26h 34m 0s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-64-916464-2/iSCSI/ : SUCCESS in 23h 24m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"d8d04b83381c32a8ea15b2418ccec6931a8da4fd","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-22 13:23:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/2/20240422133017/NEC-ISCSI : SUCCESS in 4h 11m 55s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/2/20240422133017/NEC-FC : SUCCESS in 8h 52m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d435ec3bcc1c3b41f465f2cfe890e07f9894aef1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-04-22 20:36:47.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/2/20240422180137/NEC-V-ISCSI : SUCCESS in 6h 47m 24s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/2/20240422180137/NEC-V-FC : FAILURE in 11h 35m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6a078c07199b55b4d983ce952cc797e32822643f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-23 04:55:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/258/.","accounts_in_message":[],"_revision_number":2},{"id":"1a1aec54032af08c00604e92cf6dd623ebc93533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f4efbe26bb2289ebcb1e8cac4583ec84628f7cd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:22:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/104fe749d8004b19aa98141f7ca17be2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8ce4b6b434a4edfa3276c6fd232ba23 : SUCCESS in 26m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9bfcb93f07a34ba1bc392f463a66a12c : SUCCESS in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f19c29f60b1907375f2f97d42cef154ddf5014e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 14:14:52.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"d48e888c033bfeba9450ebfa1a4ecdd6e47991d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:25:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/847262b5dedf40e3946858d60e8cf981\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a2c3aaa61a741a0ac70e67dae3ae8f4 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aae86b9bc8b435a86316da72f3891fc : SUCCESS in 12m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cab8c778000e425598d5a9347fb402c9 : SUCCESS in 11m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16107d1f794e4464bf09115afd228f74 : FAILURE in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3217c46ef4824a2e968a5c91633998b6 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/304257e6aeaa4df18c63d4b4c3816021 : SUCCESS in 1h 04m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd068499eb124114a445dae968909e92 : FAILURE in 1h 44m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13abc8871d7d4b968fa80de99b0e095e : SUCCESS in 1h 06m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e7a39e9025e44af0874b07dfc03fb470 : SUCCESS in 16m 12s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e2dacb1ed0de470b99d082c2dbf68c25 : SUCCESS in 12m 06s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9066694f45a34446857c833b322b3cb2 : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/882a5df4bfbb47f49c7fafe5cd887bbb : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b813e31698940919a7b90491f021242 : FAILURE in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d451420d54974a358a0d2f5da02d5374 : FAILURE in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b605733bc484579b87ab5d9ea941f64 : SUCCESS in 47m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0db654be77cb4a2c920d6e289724ff74 : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/73af9fd6d43444d780c4c6da49b1bc5e : SUCCESS in 1h 51m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d665b19d352b460dbc28ca6ec35180a7 : SUCCESS in 2h 21m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c06b6178bcb54b7e83ad792c2d3702dd : SUCCESS in 1h 45m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/231360eb76c644208625552f877351c5 : SUCCESS in 1h 49m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b6fc92884b4048dd9abc21287440b057 : SUCCESS in 39m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3831ba86d7724ec4931129338082f428 : SUCCESS in 1h 11m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f9897aad918b45c68b3f8c9c520ddb40 : SUCCESS in 1h 37m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c44bce0813b84142a7729b2d94271363 : FAILURE in 58m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c947e8a36f7e49e994c53f608b522530 : SUCCESS in 1h 21m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/76334c2b80a640a5afc3448a30a68546 : SUCCESS in 1h 16m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cdc4b2edecf049bea1a7e16a96705ec5 : FAILURE in 1h 30m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80b68219a86e44fe974c95df59eb04b8 : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":3},{"id":"e8d6e93efba8776e1d7c467965711cd16e00c1c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 16:30:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916464 : FAILURE in 13m 06s","accounts_in_message":[],"_revision_number":3},{"id":"0273fb2a10fb944bff22ce476c4cf662557f763e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-24 05:16:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/64/916464/3/check/fujitsu-eternus-dx-iscsi/615d204 : SUCCESS in 3h 42m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/64/916464/3/check/fujitsu-eternus-dx-fc/f2fcefd : FAILURE in 23s","accounts_in_message":[],"_revision_number":3},{"id":"df388494258189de81d0881c8f97fe3acfae3a71","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-24 12:17:01.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/64/916464/3/check/DellEMC_Unity_FC/cba7470/DellEMC_Unity_FC/1751 : SUCCESS in 1h 22m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/3/check/DellEMC_Unity_iSCSI/cd410b6/DellEMC_Unity_iSCSI/1781 : FAILURE in 1h 55m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6db3737c7a1a494354c27169c5c10e5ff95353ae","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-24 22:42:06.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-64-916464-3/FC/ : FAILURE in 22h 54m 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-64-916464-3/iSCSI/ : SUCCESS in 15h 43m 48s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"cde3e7386ad96e4c5891566ea8c1ac3a4bbeaa41","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-25 03:12:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/291/.","accounts_in_message":[],"_revision_number":3},{"id":"b7cfac90ff6fc05c27798a979b9d64b7eefee635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-25 13:02:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"97c0e78872f7e07320d14383bc9a564a4bbe5470","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:25:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff15d653ee34a3495b2762a9c427663\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b301455366c841ce96c00552d5e917df : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/97d6bfef243c4dabb4004a27863a2645 : FAILURE in 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f12161fa33fef56b364f18e1f8c5a09898a18477","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 13:49:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916464 : FAILURE in 23m 16s","accounts_in_message":[],"_revision_number":4},{"id":"546a09efb0e6a3d6d82e4f19a398828a42b8c4e5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 14:14:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916464 : FAILURE in 19m 38s","accounts_in_message":[],"_revision_number":4},{"id":"556a115394ae132c17a02070167eedecab87da52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:17:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d7357f40651498fa47c64755a8ed3d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83de8c3e6a78453e9a28c643ee56bc86 : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7656e4ef1b9b44de9e0ff9ed97a67189 : SUCCESS in 13m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e182695644374cb79fcda2d3e90d7b1c : FAILURE in 18m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4521e5d6d00f43feb6d78ecd828b489d : FAILURE in 12m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5670a568d55c4e73aaca3b1da01165db : SUCCESS in 16m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/9767d4956282451193daa560e9c4f512 : SUCCESS in 1h 02m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28c8d0f923724897b0545f21321dec07 : SUCCESS in 1h 16m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b515164214d4cbfb0175acf6c2e4254 : SUCCESS in 57m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e8cc19e347a464f82ea4e182a776a9b : SUCCESS in 14m 16s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2e028770ec08483a939ba72ffa8c9cfa : SUCCESS in 13m 39s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ef4d440428254e21966bfbcc9450e865 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2401278d8dac410b95fed84ad4b7e220 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40928a01f0d24d7189d19b929d3d19ac : SUCCESS in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d8f87fe2b6444b72a0d08fa17ef5d0f4 : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e39b2e1c055a4bddbff6db649d515e8d : SUCCESS in 34m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc726a85609a4326a34eda817d5cc200 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f11a875752e47f08eb9844d7e9fdb93 : SUCCESS in 2h 08m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bc0781a99e1c4095ad19b1af812bea52 : FAILURE in 2h 00m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/203cc6d9fd144b218aa3d0f71b915bcd : SUCCESS in 2h 02m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/20f188c51f564b3a8a62c4e8c2a0efe6 : SUCCESS in 1h 38m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/99ffbe031bd149848fa27b06460d1887 : SUCCESS in 28m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7e362de515964f708f46f29420ef4126 : SUCCESS in 1h 08m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/88662f4ad1794a82a4f58f2c562054a9 : SUCCESS in 1h 22m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1d9a8dec62894affafc0f387f8859356 : FAILURE in 53m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8f54a87025cc4e4fae40861c911ded17 : SUCCESS in 1h 22m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/448de7bce4dd4ecba0222f6ee69285d2 : SUCCESS in 1h 32m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c2d7b176d744fd9a5de18c00efada2a : SUCCESS in 2h 05m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38e9108e54c64e7bbe40d685bea6b6e2 : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":4},{"id":"a8d8d1b216d1017da6bed4faf14694f1b85ab580","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-25 15:55:54.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9aaa7314fbe14424821873a634c23ff8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0b4ed1aadb3d4e61a74c6a7c9209b5de : FAILURE in 1h 29m 35s","accounts_in_message":[],"_revision_number":4},{"id":"112d80c1681e7fcf4848269e22b886fc80b3ef10","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-25 16:31:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/64/916464/4/check/fujitsu-eternus-dx-iscsi/71ed2a0 : SUCCESS in 3h 25m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/64/916464/4/check/fujitsu-eternus-dx-fc/bafc060 : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":4},{"id":"5534bd6c1a934f0ab77550fbd0435a0105eb3f92","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-26 01:21:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_Unity_FC/40589d2/DellEMC_Unity_FC/1772 : FAILURE in 1h 32m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_Unity_iSCSI/544ed5f/DellEMC_Unity_iSCSI/1801 : FAILURE in 1h 50m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"7a5a8d181ea17ff49b81da302169fa90922faa5f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-26 06:12:06.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/310/.","accounts_in_message":[],"_revision_number":4},{"id":"725e3ef038f4c92b8e3ab3c051db4fb02cc1b83d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-26 13:00:14.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/916464/4/fc : SUCCESS in 0h 55m 45s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/916464/4/iscsi : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d32eda0255e7378ab871bc35316f4e4f3d947d3e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-26 16:01:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ddc90f413394a954e68539b4f4450879cbe3317","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:23:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23fbffb8dd8d4368b794a9218f1e63f9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70b86b0c283845aaa1cfc176edeae64a : SUCCESS in 16m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/81eaef2268c8401ea72b18537c4950e4 : SUCCESS in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ebe457e6e52408b052a21453719bd26f8af643a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 18:42:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b8b2e1b464544d0bfece5f7031f7f64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48cbda7d28364caa8beb648b10927b25 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ba3f6e685bd46ec831639194d80584d : SUCCESS in 11m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3a91135a87240738e0205f106969fce : SUCCESS in 11m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d30ecee03494fea9673e61fe5ee5780 : FAILURE in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb1cb6c9f18a4c0da7efad4f5d8afda5 : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/f692e03942c84a58b230ce89c3315872 : SUCCESS in 1h 06m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53f4b824bcab43ab8f12dcac3d0ee0e1 : SUCCESS in 1h 29m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/468175d8d1364884a237b7c767ce1e07 : SUCCESS in 1h 02m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c7f40d266c944f89b952f19570358ea7 : SUCCESS in 16m 29s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3213953629314353a2afa026f4735f41 : SUCCESS in 14m 04s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c819bb7fd084355a19f8cccd1265921 : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98b6605aded44357a4f5691b11c0d67e : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69cc3a2f2b864ae1b06b73a1c60f9f1e : SUCCESS in 7m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/55ab20b984154c5eb5cd369232d8956a : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cee15121d3c24dc3b4ed85fbafe46aba : SUCCESS in 35m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1674494f93e44e3c8b3fa549c33b7c2f : SUCCESS in 7m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b260509d046e4e58ae379bcf8ea781d0 : SUCCESS in 1h 38m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4eee23444784493a969ae48ea7811232 : SUCCESS in 2h 32m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/69676e27a87b48c2b8eaeb74d708c1c1 : SUCCESS in 1h 41m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c25b22c069e7431690b3f5c719053b38 : SUCCESS in 1h 49m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/74c5810df1b94e068d40080d4d283f48 : SUCCESS in 31m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e0267f519a0440509c652b13de824eec : SUCCESS in 1h 11m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fffefdad6cf24b299549c154c7b8c8f9 : SUCCESS in 1h 45m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/408e658add54451f8e8c35daa8f899ff : SUCCESS in 46m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ae17ac53d4ab4cbaa9ef24b9967445e6 : SUCCESS in 1h 28m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e221ec90a17945bdb96bba8565944a27 : SUCCESS in 1h 47m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e76c8401d3464defa19c42e70ce52113 : SUCCESS in 1h 39m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dbb87f0a00543988cda819aed09005b : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":4},{"id":"5dbc503b615a6342da9e7a68e63a36a2d3c0ca32","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-26 18:49:30.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_04_2024_18_43_23\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"cdcf9a49185a6d6f560a43f5b407588c12cb954e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-26 19:08:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/4/upstream-check/cinder-tempest-plugin-ontap-fcp/18f3c2d/ : RETRY_LIMIT in 4m 03s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/c9bbb76/ : RETRY_LIMIT in 3m 53s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/4/upstream-check/cinder-tempest-plugin-ontap-nfs/f1a02d6/ : RETRY_LIMIT in 3m 46s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/daf99d9/ : RETRY_LIMIT in 4m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"73fdba9f26c1f2f4c2e427d442e424f7ea48db12","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-26 22:02:26.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a720e9becd3e4dde9f3154dc8dc07fb1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e618533108664f5ca993745b1c739bc5 : FAILURE in 34m 11s","accounts_in_message":[],"_revision_number":4},{"id":"04331592983c28b7de966b92bdb6dd566b945083","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-26 22:07:57.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_PowerStore_iSCSI/8fbf572/DellEMC_PowerStore_iSCSI/1282 : SUCCESS in 1h 13m 34s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_PowerStore_FC/e1c0e27/DellEMC_PowerStore_FC/1453 : SUCCESS in 1h 05m 20s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_PowerStore_NFS/f25e142/DellEMC_PowerStore_NFS/972 : SUCCESS in 1h 15m 29s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_PowerStore_NVMeTCP/33dbec4/DellEMC_PowerStore_NVMeTCP/833 : SUCCESS in 1h 44m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"2649f28029b396a2a5ee97e6c9503e041ca4d77a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-26 22:36:34.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_Unity_FC/c448d40/DellEMC_Unity_FC/1784 : FAILURE in 1h 30m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_Unity_iSCSI/2505066/DellEMC_Unity_iSCSI/1812 : FAILURE in 1h 57m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e342ee7ef072a53ca6c97d4810010071d06624f7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-04-28 01:59:54.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/4/20240427232817/NEC-V-ISCSI : SUCCESS in 9h 09m 49s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/4/20240427232817/NEC-V-FC : FAILURE in 11h 31m 34s","accounts_in_message":[],"_revision_number":4},{"id":"83a93775fb547f04b47613616daed3ba6c68b380","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-29 10:46:07.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"ce7bdf5b17b53e0dd79d1c41e3e4a3d12e388c6c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-04 18:16:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/4/20240504222729/NEC-ISCSI : SUCCESS in 4h 22m 43s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/4/20240504222729/NEC-FC : SUCCESS in 4h 49m 02s","accounts_in_message":[],"_revision_number":4},{"id":"7d3019af93f8a5ae6b8653e20eff583fe6a45bbd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-10 20:15:03.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_PowerFlex_3.6/d1b56b9/DellEMC_PowerFlex_3.6/1272 : SUCCESS in 1h 47m 32s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_PowerFlex_4.5/968d6ee/DellEMC_PowerFlex_4.5/206 : SUCCESS in 1h 54m 50s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/64/916464/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/c4372ad/DellEMC_PowerFlex_4.x_NVMeTCP/147 : FAILURE in 2h 54m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"ec10d90e0fca73ab2c4a36e00a4c6628898756ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6d964705ebfbbfa1cd64820e8986cb65d2f10180","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:17:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b111f5d924324d8daeb1b63141e36b60\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e1b911f3c8c485d8bb25a5163a0137c : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc2bd26fe8564409b478b4207d096665 : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c847e6552ac965865b88a1c71607ffd84836be54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 18:12:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59244e06eb8745189583c175f5f09ff9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82eea9eaa3c242b1adf2cfc786da34d5 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98fe818bf245430eb27719d9e69539d5 : SUCCESS in 13m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/259bd703b0264fa497be0fdc2a6d813f : SUCCESS in 10m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75d2b648a2024754a1983656ed0f14af : FAILURE in 10m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651a147d1183432383ff214fdcc4f630 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/800f35112dbb4946ab740c83fb9b9a0c : SUCCESS in 1h 01m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2bbc8511def4baf83ab2421d3d9bdd6 : SUCCESS in 1h 15m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42c5e0f21dab4f638a5c7769cb28c583 : SUCCESS in 53m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/940ebdc4bdcc45ef98a0fd48a120d194 : SUCCESS in 17m 33s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1a16847035e04bce9e1d5c7ef47c3302 : SUCCESS in 13m 38s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d836dfce701f4852bd97c08eb50cec7d : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e7ec97cea68347b887388e406f50faaa : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6ca5b05bc0d4f22abe48c49ef1d3ed6 : SUCCESS in 6m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3da1507ba56f4714b83f0b11713fbdc9 : SUCCESS in 9m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/280eff6e9c024e388a742250ab73c1df : SUCCESS in 55m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/834c997e440342cba979f4e330ef5d32 : SUCCESS in 8m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a0bec5d39ea4bbf9707759b48b48416 : SUCCESS in 2h 10m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/168a370eddeb48078767ab03ed5f2065 : SUCCESS in 2h 49m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/35d53469050d4e70b3a1f622839a43f6 : SUCCESS in 1h 57m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5bf81c4447274692b08c36c9359c4557 : SUCCESS in 1h 27m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f72e9d5306b1459d98b20e1ac8569175 : SUCCESS in 38m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad13f15fd82648c4b9727d2a96c930ba : SUCCESS in 1h 11m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7489c6a3bcd8490b9fc98e91573ce1e8 : SUCCESS in 1h 38m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4dfe5fff8ffc45ce973c13b717ea15d4 : SUCCESS in 47m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8e85e2c98d2d47e5ba1c8ae209279ac3 : SUCCESS in 1h 33m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8055a70ac2614e9aa9027a4db1a13686 : SUCCESS in 1h 12m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72fbe92fbac04a939499f09e6dc1d389 : SUCCESS in 1h 27m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f03e3efab82485cbecf91e9dc2c6c23 : SUCCESS in 1h 12m 33s","accounts_in_message":[],"_revision_number":5},{"id":"2eac152088a84edc4cb8f6dca2b3dee5623576e9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-13 18:34:21.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/5/upstream-check/cinder-tempest-plugin-ontap-fcp/4f3cdce/ : RETRY_LIMIT in 4m 44s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/236b303/ : RETRY_LIMIT in 4m 39s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/5/upstream-check/cinder-tempest-plugin-ontap-nfs/4b18c93/ : RETRY_LIMIT in 4m 03s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/328caea/ : RETRY_LIMIT in 4m 09s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"80731a4269139fbbd4084787e76e2337df142220","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-14 00:56:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916464 : FAILURE in 22m 40s","accounts_in_message":[],"_revision_number":5},{"id":"e13ca7b1326e48be22fc1d21790291d5d5a2e843","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-14 02:14:31.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1cf7c7fee19649feac77da983c978a6a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4c4e0b23faf34c29b23ff9fea4cbbbef : FAILURE in 1h 30m 17s","accounts_in_message":[],"_revision_number":5},{"id":"844e5604a23c6c807ccd969af2cee0b578e4fd2a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-14 11:30:43.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/375/.","accounts_in_message":[],"_revision_number":5},{"id":"507c95eb94d891d32ac9336ba01c7d1afe612229","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-14 18:28:44.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/64/916464/5/check/DellEMC_Unity_FC/0c2dd46/DellEMC_Unity_FC/1854 : FAILURE in 1h 47m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/5/check/DellEMC_Unity_iSCSI/500bbd1/DellEMC_Unity_iSCSI/1882 : FAILURE in 1h 52m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"34ec90033ae592cba754c8365db87a16460db11a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-15 07:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/916464/5/fc : SUCCESS in 0h 39m 23s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/916464/5/iscsi : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":5},{"id":"c12d93c349472d5bd9f1be97e0d1eefa5809401e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 08:17:51.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/64/916464/5/check/DellEMC_PowerFlex_3.6/0faffbf/DellEMC_PowerFlex_3.6/1336 : SUCCESS in 1h 33m 18s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/64/916464/5/check/DellEMC_PowerFlex_4.5/f0492d6/DellEMC_PowerFlex_4.5/261 : FAILURE in 56s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/64/916464/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/a311dc3/DellEMC_PowerFlex_4.x_NVMeTCP/208 : FAILURE in 2h 54m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"34509d9dcf007a329d7a0882f352a221e6da8fd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"982c217a4a6387a1365a302d4d291f65c83214cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"9245e98be42de67e49bb3042e390dacd695ce273","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:31:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ddaff9cff34a12b347d282a7d90e41\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/552002fc9e774e4984b8181c66082b26 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0d412ec4f1d0417987c8176c658dca3d : SUCCESS in 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"50baf145603be0a1b3fbd6fa606575e612a7efa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:26:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3c7681e286c4c65b1c9d2114407f679\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5f801bd156459a9e6f07968ca9a086 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70e362206a5744458370e7aa62fd74e6 : SUCCESS in 10m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9afbe576c540490fae6885db540971ac : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9820a50df1cb4534bb90b8c9ee2e36ea : FAILURE in 14m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d590bef46745c6a6a8249130087594 : SUCCESS in 14m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/c8c65a17e1b7489eb329958a69d19431 : SUCCESS in 1h 07m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/82d19786f6bf4ea0ac0bc23864b5eadb : SUCCESS in 1h 29m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7784a671bc34711938a78830f5c3946 : SUCCESS in 1h 12m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/91024eb38a244178b34f3a2acf138687 : SUCCESS in 17m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ce3dfd52322d46ab8c22783140478803 : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce79712059344298836f87da0add78d7 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc7c6321613e4185bebd8f73f689df5d : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/40974920ddfc4be4846df36fb13d90f3 : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c1b3506bb6424850908f9d7676a433b0 : FAILURE in 48m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ddf3ef01bb8d4dae975e332125aae381 : FAILURE in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/37ef2f24bbc9459ba1c87cee64ab378d : SUCCESS in 2h 03m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0cd0b6e166d74a8a83b9d35315fc201e : FAILURE in 1h 58m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/962229e513b74eabafe32b5aa9d6597a : SUCCESS in 2h 00m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2087149b61bd4d0e96fb2f1cadafbff5 : SUCCESS in 2h 04m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e346b19f75a1408b9775cf03e39d2b78 : SUCCESS in 28m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/db598e15288b459d9278799355014c9a : SUCCESS in 59m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6d1e56c4f2db499bbe0197d2efa16e1e : SUCCESS in 1h 08m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/65e0220057bc4b5f9b0336c0072518e8 : SUCCESS in 50m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f5a8743658c7451b85dc8389b1a6c92b : SUCCESS in 1h 00m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1af4e23dd73d4f51bef3b1e468914d12 : SUCCESS in 1h 25m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4d5a6a6d1214f9e86bffad31d2ba673 : SUCCESS in 1h 30m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5045c47eb2114dbcaeacaee6b7751007 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":7},{"id":"fd2e3efcde5845f9a58a789a4ae73955b299494e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-20 12:43:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/7/upstream-check/cinder-tempest-plugin-ontap-fcp/a5a1b36/ : RETRY_LIMIT in 6m 54s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/cdf5fcd/ : RETRY_LIMIT in 5m 50s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/7/upstream-check/cinder-tempest-plugin-ontap-nfs/24af985/ : RETRY_LIMIT in 4m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d3ff612/ : RETRY_LIMIT in 5m 29s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"77e59404b21fe1daf0e05a5570a9bca2bf9924e5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 15:31:13.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b0a607f3aed44790ababf7dc68779f66\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/23855350bc7448b7b0866a41fdfafecd : POST_FAILURE in 14m 16s","accounts_in_message":[],"_revision_number":7},{"id":"eb2fa5d2d1c587f1e5234f1a5ee33377835f801c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 16:22:59.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916464 : FAILURE in 23m 38s","accounts_in_message":[],"_revision_number":7},{"id":"909756c16d0008d0fe3bc724069b54cc5e85b7af","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 20:21:46.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916464 : FAILURE in 36m 41s","accounts_in_message":[],"_revision_number":7},{"id":"58287eb4e2dd16565ab82e9ac2af392c55271c79","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-21 05:03:50.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/7/check/DellEMC_PowerStore_iSCSI/5305f10/DellEMC_PowerStore_iSCSI/1350 : SUCCESS in 1h 43m 41s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/64/916464/7/check/DellEMC_PowerStore_FC/b1904de/DellEMC_PowerStore_FC/1533 : SUCCESS in 56m 26s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/64/916464/7/check/DellEMC_PowerStore_NFS/48e1be2/DellEMC_PowerStore_NFS/1044 : SUCCESS in 1h 04m 46s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/64/916464/7/check/DellEMC_PowerStore_NVMeTCP/bb2d76e/DellEMC_PowerStore_NVMeTCP/898 : SUCCESS in 1h 35m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"8025c7e7ab98de9b0193ca323ee3cc60e28a0483","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-21 10:03:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/438/.","accounts_in_message":[],"_revision_number":7},{"id":"596236ea7d27d3080bf35b03f679a21a978d9b00","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-21 12:00:07.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/916464/7/fc : SUCCESS in 0h 42m 42s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/64/916464/7/iscsi : SUCCESS in 0h 51m 16s","accounts_in_message":[],"_revision_number":7},{"id":"aedb4d202730043ce6d3d8604a75e37b35f1cae9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-21 16:35:07.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-64-916464-6/FC/ : FAILURE in 19h 41m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-64-916464-6/iSCSI/ : FAILURE in 17h 31m 45s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"8d7afcf1986773b109ff2f35843c3e163ca4563f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-21 19:00:58.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/64/916464/7/check/DellEMC_Unity_FC/e80b182/DellEMC_Unity_FC/1914 : FAILURE in 2h 00m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/7/check/DellEMC_Unity_iSCSI/bceb004/DellEMC_Unity_iSCSI/1942 : FAILURE in 1h 56m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"caa06a80cc534a6fde4411362017f22f761f4015","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-24 02:57:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/7/20240524070531/NEC-ISCSI : SUCCESS in 4h 26m 49s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/7/20240524070531/NEC-FC : SUCCESS in 4h 51m 48s","accounts_in_message":[],"_revision_number":7},{"id":"7a84f8abd637ccc28ec4215d44e751d135908281","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-24 21:38:57.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/7/20240524223739/NEC-V-ISCSI : FAILURE in 7h 05m 02s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/64/916464/7/20240524223739/NEC-V-FC : FAILURE in 8h 01m 15s","accounts_in_message":[],"_revision_number":7},{"id":"8ddaddc1a31699a92bf8eb9017ec8c2c15e56aac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e853f27068ac99612d18790bd3170f31f2465319","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:42.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/592/","accounts_in_message":[],"_revision_number":8},{"id":"1ca91b4c79add1f7dd4d5088006431ba166875e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:44.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916464/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916464/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"ca44b595e5b8a3a2b4ac1e045b1eb0574ec87dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 17:54:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db529de0a31e4a1bae6eed21b2d839c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8bbeb66231488c996fb62a88b61538 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d320b2094b804339ad1d6749a19c4887 : SUCCESS in 12m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98f487e039864d40b7a7bf54c1beac4c : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6ada56e780b42ad9948ff00d84154ea : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca831e8d606a42878197af58d8bafa70 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/481ee0512e674cfe8a4d236001e78211 : SUCCESS in 56m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab0d313e2cd24dcd998bc1224e2ef006 : SUCCESS in 1h 39m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/486a879d5ad443feaf7ac1a92ebcb9ec : SUCCESS in 34m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7b305f4edbf34475932f518dba610987 : SUCCESS in 24m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8a7867b0424d4a8da16b1d3d845a566e : SUCCESS in 12m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8503def9f440456989a0004e6bce757d : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5eb1fe31be084e17ab9cfed2ddcdcbfd : SUCCESS in 12m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fdff5bbf8f7f461b9277b6bbbe2e9cd8 : SUCCESS in 8m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2c1e7bf906ee491787da76952c53c4f2 : FAILURE in 35m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e79652b3a7d042348a195928c3ec8e87 : SUCCESS in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd0c2f4f3fda42dc8eca69b73c06c420 : SUCCESS in 1h 58m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/763f4e1cb69743fa922bd41d6be6370c : FAILURE in 1h 42m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c6d0fe5f3adf46da97d5c36fdfdeca00 : SUCCESS in 2h 09m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f83daf44bbc34a128a2f456ec53350b0 : SUCCESS in 2h 00m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3d486cb2feca4fc399c775bb677d1982 : SUCCESS in 34m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c1b961ad552746b79cb4f53d35b46e0a : SUCCESS in 1h 14m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/96e073250df04478bd8ad1ce3aab1239 : SUCCESS in 1h 37m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b367af7c92564a1096261d74b0ce90ca : SUCCESS in 34m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d9d8f9d627fc41b8a72600c7a57b5ec3 : SUCCESS in 1h 29m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b615ea15a7704471bc1d0263b3e741d2 : POST_FAILURE in 2h 24m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e1c0ea6cb5694095a92edad641481e51 : SUCCESS in 1h 42m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3019e8c776a04ce69b872cd0c9974a95 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":8},{"id":"0739de2e610247168842420f9f59777be5df6827","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-18 18:01:19.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2024_17_56_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"64904a9fffd0a4aa26f7996e2ac17283bfdee718","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 19:11:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/8/upstream-check/cinder-tempest-plugin-ontap-nfs/c370db7/ : RETRY_LIMIT in 20m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"591d00a8764267a17e127bb7da7b9982b606620e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 19:33:40.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f83a9a154c6f40d8a0d60f9a89646755\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8735eec065b7441f9f2b5fcefd26dcd1 : SUCCESS in 43m 37s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb0ea341aff44178869784cd0576664c : FAILURE in 39m 14s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/13a535ca5ed349c585e3475766fffe57 : FAILURE in 2h 10m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9cdd756d45ce413fbe7f718d48b91c57 : SUCCESS in 47m 36s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"51ceed92847f174b9fc690ee09839d1fe33a0f19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 20:50:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/617c88ec1d624e2dafbdf577b222c97b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b382059c4094584bd21abb760b48507 : FAILURE in 31m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a28b5627767d4f1a879bfc6ea28daa7c : SUCCESS in 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0d0d2d2c89991bbdefde2407d0d54dd02ff6775e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 22:24:56.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/916464/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/d2078ad/ : RETRY_LIMIT in 26m 52s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"b578177f58638fa2e11a17b2b44945c838403e23","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-18 23:25:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916464 : FAILURE in 32m 15s","accounts_in_message":[],"_revision_number":8},{"id":"6f273ab447fe1d116e79c9327cd3752aa2b5691b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-19 03:30:39.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916464 : FAILURE in 18m 16s","accounts_in_message":[],"_revision_number":8},{"id":"18554156e5db5bf5fe549a5541c790ede112bbf3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-19 04:44:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916464 : FAILURE in 16m 58s","accounts_in_message":[],"_revision_number":8},{"id":"78901eb9b4caf7355d0e6ba0a8ea6a0100db3678","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-19 08:50:54.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_PowerStore_iSCSI/6d5f794/DellEMC_PowerStore_iSCSI/1753 : SUCCESS in 1h 29m 17s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_PowerStore_FC/695096e/DellEMC_PowerStore_FC/1951 : SUCCESS in 1h 21m 22s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_PowerStore_NFS/d7ae8ac/DellEMC_PowerStore_NFS/1446 : SUCCESS in 1h 14m 35s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_PowerStore_NVMeTCP/ac7e59e/DellEMC_PowerStore_NVMeTCP/1306 : SUCCESS in 2h 25m 18s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_PowerStore_NVMeFC/08fb837/DellEMC_PowerStore_NVMeFC/270 : SUCCESS in 1h 08m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"a07a1342c2c87af2f1622cf611fc52bae1422916","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-19 13:06:59.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3a1b2b8e9ec84c4a9923a2cdb7abaffc : FAILURE in 1h 15m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/540ed13e7ebb46c984c657ca13954454 : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":8},{"id":"3217b17bbc849a46ebc114029a425b7f3ddca768","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 02:21:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/567/.","accounts_in_message":[],"_revision_number":8},{"id":"c142e8817dd479cfeb8104a3d876feb514474c2d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 13:24:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1155/.","accounts_in_message":[],"_revision_number":8},{"id":"1db9b0a1b15dbce2f4faac8a9e77bc5e15cdf6b7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-21 05:50:45.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_Unity_FC/197ce65/DellEMC_Unity_FC/2634 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_Unity_iSCSI/ed5d2aa/DellEMC_Unity_iSCSI/2643 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"0b8eedd15307b92aca36afb5ae6b34cac527e844","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-22 23:27:50.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_PowerFlex_3.6/a5615fa/DellEMC_PowerFlex_3.6/2128 : SUCCESS in 2h 15m 33s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_PowerFlex_4.x/74a5860/DellEMC_PowerFlex_4.x/1012 : SUCCESS in 1h 47m 52s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/64/916464/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/ae3f8cd/DellEMC_PowerFlex_4.x_NVMeTCP/952 : FAILURE in 5h 53m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"0b509f789e21f3a4ddbc4ea58f4886056ffb62e3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-23 11:48:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/64/916464/8/20240923162636/NEC-V-FC : SUCCESS in 3h 47m 17s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/64/916464/8/20240923162636/NEC-V-ISCSI : SUCCESS in 4h 22m 02s","accounts_in_message":[],"_revision_number":8},{"id":"e777fef57c3a5110145c4a961b413de2d2362ea8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-12 08:55:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/916464-8/HPMSA-iSCSI/ : FAILURE in 3:30:40\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/916464-8/HPMSA-FC/ : FAILURE in 44:30.83","accounts_in_message":[],"_revision_number":8},{"id":"d13e439e355b1c77b79ced2b8f3d19bc0269a2d1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-17 12:22:09.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/64/916464/8/20250217195438/NEC-ISCSI : FAILURE in 52m 28s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/64/916464/8/20250217195438/NEC-FC : FAILURE in 1h 27m 28s","accounts_in_message":[],"_revision_number":8},{"id":"96c10b44d9dc070e57b2ae729be968aede0390aa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-10 16:43:40.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e292c230da09604ea411024a56cf9903d2d61a1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"c7aae11a7319da159c1e30ee0568f31c4db5e4d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"5b683fc96d75667bdf1ace8529db82b4c9bd82de","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 12:53:56.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"37a9dca259d4ebfab4243848664848a2219b0cb3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:54:30.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"0a11d02cbcd9ebedffae57b9488f9d74d9ab95ed","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 12:54:32.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"483586ab69631a3f36e3f3c51b13df152d893f80","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:15:11.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"8645a330cd599105c48af66df02727f6dacfa957","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 14:48:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/124b4554cc4a4bf89562cacda0753bcb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78cfbbeb84f94c5ca7ec5c00e1b2a6cf : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bfe2a2882bd7473996db5e7ba2c8ed89 : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/316864692a7f4839bc6ea957b6b2f847 : SUCCESS in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"041f487cfd93cff491a598be26ecc5f52a9b4b66","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 14:55:16.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4ab65288641638fc489bf2b1d293945c099522ec .","accounts_in_message":[],"_revision_number":9},{"id":"1b5d14058ff1b919a9f7214768f5883fe59242bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 16:05:10.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c84864047ec4aab8bae8c316c8eb103\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1ccf22b5f8423a9460f9a69050b8bf : SUCCESS in 7m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27f3be6d673840a5b87b1e79afe6be5b : SUCCESS in 12m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84825f8d63994415948f6f892569bbeb : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b642483c343647aeb67aec926fa02e6a : SUCCESS in 15m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/6b2cbaa567394547ba5db37db0aa483d : SUCCESS in 54m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c1ad58dfa884f67aae1bea5a73d832a : SUCCESS in 54m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8655184763e454d94fb20a71883de7e : SUCCESS in 1h 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/374bb47e64d445d5b75a144a152fe256 : SUCCESS in 1h 01m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/89b299931fa14f1f9a6a67ea28e07a24 : TIMED_OUT in 41m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ada6878b7a054125a12d83be250355d0 : SUCCESS in 8m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4f5f86220a1d41ae8f8a7180c5dd5f8d : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f30a799dac942c0be56ffde658b6fa7 : SUCCESS in 6m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a8defc85f22a4816a4034b0c3785a6d6 : SUCCESS in 12m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/24cd3471ad4d415183cdf5cfecfd7c8e : FAILURE in 38m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5cb3b90d71a34b899a07f3c73662e645 : SUCCESS in 7m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a837d3279a344e1955814880d4eae94 : SUCCESS in 2h 13m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/295515d6488240f3a6615ec0ec6f9819 : TIMED_OUT in 3h 06m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d30d18a1842442e7a65d1e5a581eaea0 : SUCCESS in 2h 05m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cf5d5c5b550d44c1baa48d4311b59c40 : SUCCESS in 1h 57m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/21208dbfb6544929918975d8b09aa25d : SUCCESS in 32m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9dfb5b5f7a0b40d486241773fa1b93be : FAILURE in 59m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c90b81131f984034884fcd6247467147 : FAILURE in 1h 33m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4eda304284da4005b96455f02ccd0a23 : SUCCESS in 40m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/79fd7e92972442a4b4b1464023936016 : SUCCESS in 1h 58m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0be2f808bc6c4d499f1cc39730f8cc07 : FAILURE in 2h 18m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0584332e881f4312948e9948d7d86c1f : SUCCESS in 1h 56m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c85e85246a29470aafdba62601d90bdc : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":10},{"id":"8384c4ef94232db1257646b3395e7cc6f330fee9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 16:37:05.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4ab65288641638fc489bf2b1d293945c099522ec .","accounts_in_message":[],"_revision_number":10},{"id":"e28252c0715febd6b29aac0e216870a0a1d049da","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 01:04:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2257/.","accounts_in_message":[],"_revision_number":9},{"id":"6c39c041ef849c6fc64718447193cf41b2ec983e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 01:30:54.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2143/.","accounts_in_message":[],"_revision_number":9},{"id":"65beb6af95f2200b9261a9138e2bbba34c7272c4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-25 01:34:57.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/354b2f97f46a467ea3dd869ee25d7f09 : SUCCESS in 53m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6e0e8696f0d7459fad45b3edbef305c4 : FAILURE in 58m 56s","accounts_in_message":[],"_revision_number":10},{"id":"63a79baf4c4a406695b0da5dea85c189efd95c88","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-25 01:53:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916464 : FAILURE in 35m 28s","accounts_in_message":[],"_revision_number":10},{"id":"08bccb6669570ea89f5d0f756100cc7b292591fa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 16:24:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2275/.","accounts_in_message":[],"_revision_number":10},{"id":"8567836a31b59764d533f1b4f3d2313f5e4be593","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 17:20:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2161/.","accounts_in_message":[],"_revision_number":10},{"id":"b26e55ba8fbe95d47dfd90e1f45ebff7f823d0cd","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 15:10:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b150b3a2b25c52102642effb96143d3239e97103","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 17:08:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88f6c43d0ff24f4a92db98269e9c718f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be0dae01547c43c0ad4ce00f773eb205 : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/331bbcdb269b45e4b0da8aeac7da99de : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/205da14a13704a21aa2602b1b41a2ceb : SUCCESS in 19m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3ecc593c2c747a536091a052e99773ae0566e414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 17:54:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6924c6585393415a903496133a55848b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba5352bee7248e1975728d747d9ae5b : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b46fa8e3c1740d79e82ace467428073 : SUCCESS in 11m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bbf240c0bd447e8a3d9c149fbb08476 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f9f27a0ebe4cdf82c80ae53dbee08a : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/6ebaa15d18284f10925a47b71a489934 : SUCCESS in 1h 03m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18ce359f2003474092af29063b209a0e : SUCCESS in 28m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57b3dbee4e944be0832255f1bcf80de1 : SUCCESS in 1h 21m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d8714067aac45319f3ab002e9d30228 : SUCCESS in 1h 04m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce66425929e54e3a8edd3000df1c73b8 : TIMED_OUT in 41m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dc51edbdf0694c818139cd33194efd62 : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/13e92f7ff4584cc8b0efec5f83003076 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8bc1f9612d354dbe9b16dde546109846 : SUCCESS in 7m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c28c10d3714d40c8a65c4b4aab5aed90 : SUCCESS in 9m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/24789a6ae4864e62b75be01fff628911 : FAILURE in 33m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b69b6b193e240faa19460b33ef274c7 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1974942143154eb6a23a16b97a0b40aa : SUCCESS in 1h 59m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ad7301d777414db8b3a32f370e796a86 : FAILURE in 2h 35m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/69186a51b0de46dcab1ef59d6a6aee05 : SUCCESS in 1h 52m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/50456fd94cf9493997ff4f8942831f1e : SUCCESS in 1h 48m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d6cc27c13a4a4dd6942e142fcc0ccab6 : SUCCESS in 27m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c96c4b9b99b5499ea7e1007bb13a9c36 : SUCCESS in 1h 12m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4d0497b8eea84c839b5d5879fb470665 : SUCCESS in 2h 05m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e80a39b9968d405a925982dad672581c : SUCCESS in 44m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3f84e095653444a0949ef6afd9ea92fa : SUCCESS in 2h 09m 49s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ff4a40af930643178dff0cff44cb809f : SUCCESS in 1h 40m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba772a0af6804c95929e4fb9e44e6b54 : SUCCESS in 2h 01m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/686ae0ecd147470aad302323611a0221 : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":10},{"id":"60dc7bf435af2b4a5baf466026a7f381921c55e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-30 18:10:40.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_10_2025_18_03_53\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"823e1392b419bdfa3c11d71b5cc6b613afac5159","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-30 21:31:52.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9ebb70200f5f4bd2a476f538bfbca3d0 : FAILURE in 29m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/da4ef2b5451e4c9c878596c3aac196e4 : FAILURE in 1h 02m 59s","accounts_in_message":[],"_revision_number":10},{"id":"bc1bfd7ef09ee164a029b699834cbc24cba7391c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 10:21:10.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4787e1787491291461765cfd7aa151b2c4c16e12","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-31 15:19:04.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"8572a8c00ff8b17c0ce0bd9e5dc6b7bf3eafe141","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-07 02:28:13.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0487f18455c12c5829aad0cf70ce103e74a0418a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 02:29:03.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"e8525bc3a28101306e43a7d97b4ce0801c3873ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:23:49.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/976d2d5a6a29407cb0f2d9a41a3a164c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853d33dc1ea045ab81518e19d53a0c58 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e81ddeddaaf44495aba0cbe6df686fca : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e3f9a00ffe249c9886f1ea7623c73c3 : SUCCESS in 14m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14b8a8b353b47b3b7ce1069afea10eb : SUCCESS in 16m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b1c3e96aab74496b893136843f39a1af : SUCCESS in 1h 04m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73c8fd44d2c043b49ab54dc340ea4350 : SUCCESS in 1h 00m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59adcebb33ef462ba33331ccbf37ab8a : SUCCESS in 1h 44m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abea0751d9a640318e38cd4049f9f6cb : SUCCESS in 54m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/256dedbc2dbc47b0ad66cc66f1227251 : SUCCESS in 1h 01m 25s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b062a6497d5b4f048552349ec2ecc74f : SUCCESS in 1h 42m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/440a63b4040a4c66ab8e3ee1fb9d6b53 : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":10},{"id":"350f4ac10bbf47ef27b654fddc24aaaedff3239d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:23:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"788726d216079f64e662b109438b20806ba23db9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:24:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a87ccb2ea83b4b6386ee540465aa9bca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e57dab894f5418a8984c9a4ec52ba2c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":10},{"id":"86f2e59d683383d100addf8f413b42e0ac93b8cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-07 13:32:16.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"3d296ed3d6146cb88b59dacb54df46ac748d580b","revisions":{"85b3b7cef63f0b35ea4fdf0f6d2c12f5562a5636":{"kind":"REWORK","_number":1,"created":"2024-04-19 14:51:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/1"}}},"commit":{"parents":[{"commit":"e980f899671fde1c1d08e966eb0eca0c2f0e5ee9","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e980f899671fde1c1d08e966eb0eca0c2f0e5ee9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 14:47:48.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85b3b7cef63f0b35ea4fdf0f6d2c12f5562a5636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85b3b7cef63f0b35ea4fdf0f6d2c12f5562a5636"}]},"branch":"refs/heads/master"},"db348a3efeb27a79bfdb80be504e03650a160068":{"kind":"REWORK","_number":2,"created":"2024-04-19 17:27:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/2"}}},"commit":{"parents":[{"commit":"e980f899671fde1c1d08e966eb0eca0c2f0e5ee9","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e980f899671fde1c1d08e966eb0eca0c2f0e5ee9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 17:27:28.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db348a3efeb27a79bfdb80be504e03650a160068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db348a3efeb27a79bfdb80be504e03650a160068"}]},"branch":"refs/heads/master"},"2bc5d5f69070bb71849969080fb7743d46e151d5":{"kind":"REWORK","_number":3,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/3"}}},"commit":{"parents":[{"commit":"55a5c05a2196a5aa4accfcc8c92becee54c12461","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55a5c05a2196a5aa4accfcc8c92becee54c12461"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:51:11.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bc5d5f69070bb71849969080fb7743d46e151d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bc5d5f69070bb71849969080fb7743d46e151d5"}]},"branch":"refs/heads/master"},"6afdd9bbb018b8042f35e1a104f6f3537adc307f":{"kind":"REWORK","_number":4,"created":"2024-04-25 13:02:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/4"}}},"commit":{"parents":[{"commit":"55a5c05a2196a5aa4accfcc8c92becee54c12461","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55a5c05a2196a5aa4accfcc8c92becee54c12461"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-25 12:43:34.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6afdd9bbb018b8042f35e1a104f6f3537adc307f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6afdd9bbb018b8042f35e1a104f6f3537adc307f"}]},"branch":"refs/heads/master"},"5ba63fda049aa1f36d777998f1bad51fd9e4556f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/5"}}},"commit":{"parents":[{"commit":"79e0236568db2d89cf02c1788fde44e9fb9c3b76","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79e0236568db2d89cf02c1788fde44e9fb9c3b76"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ba63fda049aa1f36d777998f1bad51fd9e4556f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ba63fda049aa1f36d777998f1bad51fd9e4556f"}]},"branch":"refs/heads/master"},"809420d1ac6026222d56b49f28e5e9b670b20aec":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/6"}}},"commit":{"parents":[{"commit":"dcff31ce68d48fe78e0b02e8f8faa826caa1d1be","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcff31ce68d48fe78e0b02e8f8faa826caa1d1be"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:00:19.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/809420d1ac6026222d56b49f28e5e9b670b20aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/809420d1ac6026222d56b49f28e5e9b670b20aec"}]},"branch":"refs/heads/master"},"3c8acd3ff967e2487236e95b08d72b688327028f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/7"}}},"commit":{"parents":[{"commit":"064b2d618e0c5a5490e37092c41e8f2c8bc68d25","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/064b2d618e0c5a5490e37092c41e8f2c8bc68d25"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c8acd3ff967e2487236e95b08d72b688327028f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c8acd3ff967e2487236e95b08d72b688327028f"}]},"branch":"refs/heads/master"},"5c4f29320389c00ba39dfc25067a4b56ec6eec68":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/8"}}},"commit":{"parents":[{"commit":"9c2904da74a10bf1f610719d09e10676a89970b8","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c2904da74a10bf1f610719d09e10676a89970b8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 13:01:26.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c4f29320389c00ba39dfc25067a4b56ec6eec68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c4f29320389c00ba39dfc25067a4b56ec6eec68"}]},"branch":"refs/heads/master"},"6ecc53f8df97c51c791c59b974c28b8392a4a515":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/9"}}},"commit":{"parents":[{"commit":"635c563007d9215ef655a802a2cd2f8d250b1236","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/635c563007d9215ef655a802a2cd2f8d250b1236"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:47:28.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ecc53f8df97c51c791c59b974c28b8392a4a515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ecc53f8df97c51c791c59b974c28b8392a4a515"}]},"branch":"refs/heads/master"},"3d296ed3d6146cb88b59dacb54df46ac748d580b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/916464/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/916464/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/916464/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/916464/10"}}},"commit":{"parents":[{"commit":"ea720062ba36dde563cc08ffd494404a4b527b4c","subject":"api: Remove v2 prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea720062ba36dde563cc08ffd494404a4b527b4c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-20 14:07:21.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:53:11.000000000","tz":60},"subject":"api: Remove unnecessary router split","message":"api: Remove unnecessary router split\n\nThere\u0027s only one API version. Simplify things through some code motion\nand renaming.\n\nChange-Id: I4ab65288641638fc489bf2b1d293945c099522ec\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d296ed3d6146cb88b59dacb54df46ac748d580b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d296ed3d6146cb88b59dacb54df46ac748d580b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
