)]}'
{"id":"openstack%2Fcinder~916465","triplet_id":"openstack%2Fcinder~master~I1238dff20d6fcce3e3c74b446b547052f297d8b5","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-07 04:35:53.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2025-11-07 04:35:53.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-11-07 04:35:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1238dff20d6fcce3e3c74b446b547052f297d8b5","subject":"api: Add missing functools.wraps","status":"MERGED","created":"2024-04-19 14:51:42.000000000","updated":"2025-11-07 04:36:53.000000000","submitted":"2025-11-07 04:35:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916465-openapi","meta_rev_id":"7cf818c5995fefda7315e46e091b428f98a1fad1","_number":916465,"virtual_id_number":916465,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-07 04:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-24 12:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-24 12:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2025-10-24 12:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-10-31 13:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 15:17:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 19:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-20 03:26:06.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-20 03:48:12.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-20 04:54:07.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-20 21:20:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-21 21:22:10.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-22 18:08:15.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-23 00:12:18.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-04-23 07:41:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-24 16:07:36.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 14:44:30.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-04-25 15:36:41.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-25 15:38:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-25 18:41:45.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-10 22:10:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-21 14:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-06-10 21:53:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-07-03 15:33:34.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2024-09-18 14:15:51.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 18:57:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-19 00:01:46.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-19 14:34:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-12 22:57:35.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-11-15 10:29:32.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-12-29 00:45:00.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2024-12-29 01:54:07.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-24 13:56:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 13:58:50.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-11-07 02:28:42.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"39fd0b97172911bfda98a604b3e32403b607b139","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 14:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cd36edb1bee30c1753f0683d7b65f0d7d554360","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:17:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb1620e3d2ac407ab72976bf26e22463\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44b47c3de2ca44f9b15139024abd1d24 : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2a43f43a29b4604b8fa8e8dd0c052d5 : SUCCESS in 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d19fb5960770befe01960246072644f55a626c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 17:27:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a4216d412fd3fd47e0886f2474916ccb61515a89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:47:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6198474d9aa545f288d0df86771874c4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fddd27db26ce4395a717111718cf60cb : SUCCESS in 15m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a72f7f7703540e2828ac5989f63508c : SUCCESS in 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a3eaf24245611542d2f6ebf68ea48440d9a2b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 19:45:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9394523eea0f41048bfce579749523a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd4bc9ad1604ded9f67a3a33d862520 : FAILURE in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db7381f7b78f462cb29b51b9e24f9bcc : SUCCESS in 13m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/293a19a4dde54fa492b2600b87106fed : FAILURE in 13m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea973ccf54de4e2c8c1cb8774667f3d9 : FAILURE in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cc3cc4bc6b4a26bdd44cdd0ad84965 : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/bc57aa0004bc42458e9bda625c7fe3a4 : SUCCESS in 50m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aac0b5eeaa1d4ff888585368bae7aaa6 : SUCCESS in 1h 39m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a270464702f4a7a893c57e13a0e2d62 : SUCCESS in 1h 01m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9d45551af5234611ad37843eaab612fe : SUCCESS in 18m 18s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6f3c3439744041afb85b6c46a1837781 : SUCCESS in 12m 40s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4fa0b2068d414e339fa5f7cce13e1060 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f4e72faeb334ea3b0f6288cb5a51102 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/32b5b41257fe44d5a20760e73ec5f4f4 : FAILURE in 6m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2e1104a2b69b43a88fa1ac0e0a635daa : FAILURE in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3f6c4c7e0a4d4c58a85d329f213f43d4 : SUCCESS in 33m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/72fd3fae065247ccb4704a3d2271ec54 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf0f70da44df45329f01329aedcd822d : SUCCESS in 1h 40m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4c46938251f54daeb12c4165a92be2cd : FAILURE in 2h 07m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b20e4f59c6a640a3bbdaa54bd458109c : SUCCESS in 1h 40m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ec02a248bdb641ffbbf05b3096b661c2 : SUCCESS in 1h 43m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8e11a16e2c344f8aaca9876ba51646aa : SUCCESS in 32m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8355dfe4d38423baa3165231f540f37 : SUCCESS in 1h 26m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/038168586db3432d8e2835ee6278ccb4 : SUCCESS in 1h 26m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/81f13428c41348cbb74c035c7c7776f5 : SUCCESS in 40m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4aa49be33f08411ab74f452ec0e3efa1 : SUCCESS in 1h 18m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/be3fc68b9279464cbf8710d9e2ad6ce2 : SUCCESS in 1h 07m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6bd96767435c4e1c8bf95842e748db30 : SUCCESS in 1h 44m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6265655f03014f4c9b6141387d971b5c : SUCCESS in 1h 07m 11s\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/api/contrib/test_admin_actions.py\n  Comments left for invalid file cinder/tests/unit/api/v3/test_snapshot_metadata.py\n  Comments left for invalid file cinder/tests/unit/api/v3/test_volumes.py","accounts_in_message":[],"_revision_number":2},{"id":"ca8ef7f856cc942def1605cbc8c9821ea8643b40","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-20 03:26:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/65/916465/2/ : SUCCESS in 1h 20m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d411dbaf520287a649036418c86e37532ee70f48","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-20 03:48:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916465 : FAILURE in 43m 01s","accounts_in_message":[],"_revision_number":2},{"id":"28f23fdc00738bb45bbee209784ed1daca1ea780","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-20 04:54:07.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ee553522abdb418ab29f1a5948a81990\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1f9c66cef51f491fafd76c53b091067a : FAILURE in 1h 27m 37s","accounts_in_message":[],"_revision_number":2},{"id":"5cba21678a8cfcc712fddc2b0cc4b516d2b704fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-20 21:20:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/65/916465/2/check/DellEMC_Unity_FC/f002e0c/DellEMC_Unity_FC/1718 : FAILURE in 1h 33m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/2/check/DellEMC_Unity_iSCSI/4032517/DellEMC_Unity_iSCSI/1748 : FAILURE in 1h 58m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c80dbc3f63850a14363d75c4d5101d5fbb918412","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-21 21:22:10.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-916465-2/FC/ : FAILURE in 29h 5m 41s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-916465-2/iSCSI/ : SUCCESS in 26h 49m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"ac97e7fb2686de79b78178c07d1d2981f82cb3ef","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-22 18:08:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/2/20240422222307/NEC-ISCSI : SUCCESS in 4h 22m 24s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/2/20240422222307/NEC-FC : SUCCESS in 4h 45m 05s","accounts_in_message":[],"_revision_number":2},{"id":"83a6ea11964581de41aad3f1459a6f0ca894209c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-04-23 00:12:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/2/20240423053651/NEC-V-ISCSI : SUCCESS in 3h 15m 41s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/2/20240423053651/NEC-V-FC : SUCCESS in 3h 35m 24s","accounts_in_message":[],"_revision_number":2},{"id":"263feb6d69111ba457d429a239e951a4be13e9dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-23 07:41:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/260/.","accounts_in_message":[],"_revision_number":2},{"id":"56a8bc84aaeaaa521e86f4259fabdc90962c8c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fbe231a04ef82b97243ce743a1119d487311041d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:34:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36397d6b47764e4a814be0f2e0a72970\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/947d59b70c4949f685a6851a3891b7ff : SUCCESS in 19m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/071af8feb1114ce4b9b4fab8733c56b3 : SUCCESS in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b7954561265eefa960bb08ce3319776777bc81fc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 14:15:11.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"b5e5440c9011dfa9d218069053c5ccdc07500798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:24:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bdd7ec565f64fe584d1ece90b033c11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/956e9f84ad9443018614e9598ad0abfc : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/579aa2023a694bf0ae81f33827e4f3a7 : SUCCESS in 11m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34cc05447f15444e95770ee5e160866e : SUCCESS in 10m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2785ebd1ccfe41f3979caf7de3792eac : FAILURE in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e5654fe6f544577920cc489b9490b13 : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/ee3fc2df5dbc4e59bbde9e9eb3a82aa4 : SUCCESS in 47m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dea89a97ba504f779f2e7f71103c6428 : FAILURE in 1h 07m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/610645f98d1040f6a38a3ca94bca4373 : SUCCESS in 1h 07m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4885349662884b22a9e5351adca13afe : SUCCESS in 17m 12s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2733213d55d24a1a86b47697ae1ed49a : SUCCESS in 12m 52s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e5d1ed872058461599be77d4f81bb5e5 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/24a8093b6ad848958922442acac0e6f2 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ea5096dd1274d9b97ac9f5837d2b5c6 : FAILURE in 6m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ac871bf5ff8b41daa0539d9a71365d78 : FAILURE in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce32ea4c970440af8be5f757c1f66ca2 : SUCCESS in 44m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ac7f5eea4744cdaa39b8390b7061a0c : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ca179052539c44b3a1466886116b32b7 : SUCCESS in 1h 43m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/006e1e90157648de907fa75dc67bc655 : SUCCESS in 1h 55m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0516b8f5f12d469bac4c491e2aeec440 : SUCCESS in 1h 46m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d40eb84a0e2a484d87a20d2bb4c016f4 : SUCCESS in 1h 51m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6065782668d64c3992472745e4f1609b : SUCCESS in 28m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/31b800f7d00542b88886f950438efcee : SUCCESS in 1h 08m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82064092f218492a91a0d3dd4d6f9a4c : SUCCESS in 1h 26m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7bd290b90cbc4d42a0d9f36fb96daf06 : SUCCESS in 56m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7051cbaf14ec417792a063aa77af58c9 : SUCCESS in 1h 11m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/14ede01fd9f54208adbcf3bd8c334895 : SUCCESS in 1h 30m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b8cca4ab7934d8d8caffc1db105024e : SUCCESS in 1h 35m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc3a41c32a5b41b1a3fd8221a6856f00 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":3},{"id":"5e33afd4c3557bf91ff71e80d0389f12b62d5ffa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 16:49:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916465 : FAILURE in 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"b1519ee0eaaa5692fda528315d3f6135f55240f9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-24 14:13:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/65/916465/3/check/DellEMC_Unity_FC/22a5cf5/DellEMC_Unity_FC/1752 : SUCCESS in 1h 27m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/3/check/DellEMC_Unity_iSCSI/9822281/DellEMC_Unity_iSCSI/1782 : FAILURE in 1h 55m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8e9802ffac03dac4d673fd7580e6b4467c1c928d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-24 16:07:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/916465/3/check/fujitsu-eternus-dx-iscsi/99a2eb1 : SUCCESS in 3h 41m 45s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/916465/3/check/fujitsu-eternus-dx-fc/0f42f48 : FAILURE in 3h 20m 44s","accounts_in_message":[],"_revision_number":3},{"id":"e6529d95c2dba86c6131759928b59e9d3287e72d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-25 02:24:26.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-916465-3/FC/ : FAILURE in 25h 50m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-916465-3/iSCSI/ : SUCCESS in 21h 6m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"4eaa22e8bde77b376fb137b81d04a5440b0b781e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-25 04:52:26.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/292/.","accounts_in_message":[],"_revision_number":3},{"id":"2631c588c8a06fc3bba52a372f0541f609d09639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-25 13:02:12.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"32a559d2aaf45697283560cc310be0ffe7abed21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:23:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6fde38dc4194a7493c13af1a5299d2c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7fcae7a322546239820060c7c7d0c81 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee49563d863242e987b4895983f22ffb : SUCCESS in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"02cdd05843c247746f8af7bc96551e1932fe5634","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 14:44:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916465 : FAILURE in 25m 11s","accounts_in_message":[],"_revision_number":4},{"id":"b47ad3a736aa1d6cd2b52c3e79796154270a7c06","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 15:16:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916465 : FAILURE in 26m 32s","accounts_in_message":[],"_revision_number":4},{"id":"7ca089e966d05a9a16b90d0833a227b8794f8651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:30:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/390e9b85a9464ff989afffa4fc6a0951\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a510a8bc2734018a67e0e7718ecbb9d : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e5993e8297e41538ab43f1340494a3e : SUCCESS in 11m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c268acfb9d04cb584f5168c272b6665 : SUCCESS in 12m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69bc1d4c79354c2c9688d293c414d7c3 : FAILURE in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f8e856379d4fb8b02178376badbb87 : SUCCESS in 15m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/d508a45883e64536996b64c4c8ec55f1 : SUCCESS in 1h 10m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1febb35f50504a9ab6282d176a4ff3e5 : SUCCESS in 1h 26m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbb0dbfdf70e48f0bc30b30a8788fa1a : SUCCESS in 1h 04m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2fa62afaab34d4f88052a9db4440c1c : SUCCESS in 15m 37s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2f1b80c879e14827975405e43975b9b6 : SUCCESS in 15m 27s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/537682290d3a45d6a3a22b5889c39406 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5b1d534be8a445228f48cd68ce19b34b : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cdc1fd510c0d4e0986993c128ae4621c : SUCCESS in 6m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/032c9f98b9a34ca7ac6abdbef69a8c17 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae058c62a637452990ecd54e14aacb51 : SUCCESS in 38m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a15fd1b1307a428fa7fc27e6d9254001 : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/65649714ff944757b50949592f790446 : SUCCESS in 2h 03m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5a260033cd834a768268f86047903771 : FAILURE in 2h 22m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/60da3106533e4ca2b5501425415fd763 : SUCCESS in 1h 30m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e834ec2f8c22401d84b72868d18f223f : SUCCESS in 1h 53m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3856b3ae63f34d0ba4a2eb7dd7c2603c : SUCCESS in 40m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/35066e933ed145aeb2981978d6d9da77 : SUCCESS in 1h 13m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e0b32ccb59df43a49ab9b39aeaf7746a : SUCCESS in 1h 43m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/849e4efa9727429fa8723c2a0148e686 : SUCCESS in 52m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1e15ebe4e88748ca990108ca78de756e : SUCCESS in 1h 09m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7c46e8491b894b6f88390d1cdef02335 : SUCCESS in 1h 53m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aacef3fba31f4965a1877c27ec02dd2c : SUCCESS in 1h 41m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca4cde2863a84f3bb4c6a84faa020740 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":4},{"id":"e94dbbaab351ce2d30fed6e89252a657a4c4105f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-25 15:32:39.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/06e57cb864944b94be51ea0f8fbae283\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/928913f762cc486294938d2937b5a145 : FAILURE in 1h 31m 11s","accounts_in_message":[],"_revision_number":4},{"id":"39726bcba9681a7eab1a0590899714e584a71384","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-25 15:36:41.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_04_2024_15_30_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"9aa8b7cd73c420b8575c6d5f5adcd1e55e463a20","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-25 15:38:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/4/upstream-check/cinder-tempest-plugin-ontap-fcp/876c6cf/ : RETRY_LIMIT in 4m 54s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/46c0733/ : RETRY_LIMIT in 5m 12s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/4/upstream-check/cinder-tempest-plugin-ontap-nfs/339820d/ : RETRY_LIMIT in 5m 11s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/9095cf1/ : RETRY_LIMIT in 5m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"798595ea849bf416c898cc3185c36e89a25672bc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-25 18:41:45.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_PowerStore_iSCSI/4f4a935/DellEMC_PowerStore_iSCSI/1275 : SUCCESS in 1h 10m 22s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_PowerStore_FC/48f25ef/DellEMC_PowerStore_FC/1446 : SUCCESS in 57m 05s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_PowerStore_NFS/33aaf71/DellEMC_PowerStore_NFS/965 : SUCCESS in 1h 00m 44s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_PowerStore_NVMeTCP/cef6df9/DellEMC_PowerStore_NVMeTCP/826 : SUCCESS in 1h 42m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"3d9d15a1cd18c1fab7ea6ea397fccc036acf26d1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-25 19:07:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/916465/4/check/fujitsu-eternus-dx-iscsi/d6f125c : SUCCESS in 3h 04m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/916465/4/check/fujitsu-eternus-dx-fc/4748335 : FAILURE in 3h 38m 11s","accounts_in_message":[],"_revision_number":4},{"id":"81b4682829486eed7fb2a2cdc97d764de09f6ed5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-26 03:08:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/308/.","accounts_in_message":[],"_revision_number":4},{"id":"af36a8f97c70342a30b02b97b3ead51ab946666c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-26 03:16:43.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_Unity_FC/ba1d692/DellEMC_Unity_FC/1773 : FAILURE in 1h 24m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_Unity_iSCSI/4f30239/DellEMC_Unity_iSCSI/1802 : FAILURE in 1h 54m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6e593727f7a88a72bcef8ff3b17261b2ea213ee1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-29 10:45:34.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"0574f33d51e03761f1775dcfe146ddc7416d1f89","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-04 23:01:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/4/20240505031639/NEC-ISCSI : SUCCESS in 4h 18m 23s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/4/20240505031639/NEC-FC : SUCCESS in 4h 44m 36s","accounts_in_message":[],"_revision_number":4},{"id":"394e1c351d72c670bc38698ef35984fb8682d88b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-10 22:10:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_PowerFlex_3.6/d894964/DellEMC_PowerFlex_3.6/1273 : SUCCESS in 1h 39m 33s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_PowerFlex_4.5/45f016a/DellEMC_PowerFlex_4.5/207 : SUCCESS in 1h 55m 18s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/65/916465/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/6e06134/DellEMC_PowerFlex_4.x_NVMeTCP/148 : FAILURE in 2h 53m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"95cc31b143e3aea0a6056bb636268b8db4a211b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d300f6d2b0a7d6b3218bc070a4d538a8fecc1bb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:20:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f15c18ca3e74a8c8c23dc04b7da2aef\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e785bfc5731943ac9154f70d4fb00f34 : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cca216149b2d40bdbee7dd28fb4fe5af : SUCCESS in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d953a736a3ab4240348b3afc4f5f87842e34562c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:36:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ca86acad94145aca31f3526ccf308dd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3ce983fb795475c9d037c2bdb5812ac : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c5176c95edf4a6f9cbb304d7a4a3b69 : SUCCESS in 11m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/623801525c3f4b98a0f3915ebdb62bac : SUCCESS in 9m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e120c9c55044b17ba9b174431e52957 : FAILURE in 11m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc57af7ca0a047048df88c9759f8f6bc : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/4b566ac46fa04fabb7b32943760d2e8f : SUCCESS in 1h 05m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4be0b5c52084cdbbd03ca5ef497aec7 : SUCCESS in 1h 19m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de6e5f7d304a4688bf77e5bb839aa291 : SUCCESS in 1h 02m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84c1dd80f5414cf2b9db7ed6576a2768 : SUCCESS in 18m 23s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3ac1c12cb4874159958c42d189ffc891 : SUCCESS in 14m 22s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9e1284f5e22440b8b53af8902ee5ccaf : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2b85657d91f46a4994f925fa26df019 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4b675504b5a40eb8f995ee11c8ac467 : SUCCESS in 7m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1069b8d0cd214212886b3a29c73351b4 : SUCCESS in 6m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a2a31eab357847d3b2bbfd903b703bd8 : SUCCESS in 50m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e45e02698b0492596a592404be41cfe : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d6e6824948d54187a4f9ef8c6ee25441 : SUCCESS in 1h 54m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c07f867250174a568fae458cc19e8897 : SUCCESS in 2h 16m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b23ee92f62444ec5ae37e622c06327d3 : SUCCESS in 1h 33m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ca4a0b5268004ca88d33a2529d54e791 : SUCCESS in 1h 55m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0c34119dd14b4b2886bb25069967edde : SUCCESS in 40m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/187d6807fcf6421e94a5f4b1c1322b15 : SUCCESS in 1h 13m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cd3345bd5c304c2caff8ca3992eb7446 : SUCCESS in 1h 23m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/016ab68ea6884377b34a7220de6ae142 : SUCCESS in 41m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/73e4e3851ea942be8c685a31c4bf6e39 : SUCCESS in 1h 19m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e8e861277b994dd89c38182f3b3f38e5 : SUCCESS in 1h 39m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7caab055e08943b1b4f82145f7d6a144 : SUCCESS in 1h 53m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ead1a050bf2d4586af0a880208723d74 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":5},{"id":"5e30471a8ee3db128587c8180a0d771701958cac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-13 18:14:34.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/5/upstream-check/cinder-tempest-plugin-ontap-fcp/2ba6c6e/ : RETRY_LIMIT in 5m 36s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/ea8159c/ : RETRY_LIMIT in 4m 32s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/5/upstream-check/cinder-tempest-plugin-ontap-nfs/b706b93/ : RETRY_LIMIT in 4m 46s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e763e1e/ : RETRY_LIMIT in 5m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"3b71acd7acde23ceb999bd84eccd61a29fa415cf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-14 01:45:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916465 : FAILURE in 45m 00s","accounts_in_message":[],"_revision_number":5},{"id":"ce8be6f2869d9ae11b3ee96e75a8e7c510efcb7d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-14 03:02:24.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/cf9f98560b954b0ebfd5733acd9d6768\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/adb6a28875b2425d853dbc54e0d9df0b : FAILURE in 1h 30m 04s","accounts_in_message":[],"_revision_number":5},{"id":"66036b822df5130f4f71b6460fb41610a512c957","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-14 14:14:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/377/.","accounts_in_message":[],"_revision_number":5},{"id":"40ab645649f2017b272d75d38f07189d71821ab1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-14 20:34:08.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/65/916465/5/check/DellEMC_Unity_FC/ca60c1d/DellEMC_Unity_FC/1855 : FAILURE in 1h 40m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/5/check/DellEMC_Unity_iSCSI/1c837e8/DellEMC_Unity_iSCSI/1883 : FAILURE in 2h 05m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"1b39ffb87f518b5ca3a640b7aada305908921e1e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 08:18:48.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/65/916465/5/check/DellEMC_PowerFlex_3.6/317a96d/DellEMC_PowerFlex_3.6/1337 : SUCCESS in 1h 33m 27s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/65/916465/5/check/DellEMC_PowerFlex_4.5/2aa4fd1/DellEMC_PowerFlex_4.5/262 : FAILURE in 56s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/65/916465/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/849f9c8/DellEMC_PowerFlex_4.x_NVMeTCP/209 : FAILURE in 6m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"30d0ad6d2e1b0d0de407abb53852bf1245e5456f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5ddc15d32239c95e7bffec7669f2bd5edec604a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"8685deac1e512f8eb64ab276f715fe96e6437525","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:30:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e0f475c52c14bf4b9b9f23e8cd5978b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a5a73a36eef40598d4e31c7208024e3 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/43cbf17765a84166b0f98c976db9a52c : SUCCESS in 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b1fab91efcb7a7c638620a232f3d17348fe1a4ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:35:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e55e2893434cbba67214b3d5cdf7f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b113c689d284166ac75ddc9db1fbf49 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29ae86ce9d9248a18a08951c0d29f297 : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2b23563cb46438d90ecd41c43332b46 : SUCCESS in 10m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e6a8254d3494ded98ab1d22c5fdfb92 : FAILURE in 10m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da62a15293c49489b199a6d20dc5121 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/8fa7f554dba64d6db311ce1f00f52e24 : SUCCESS in 58m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6746787aec084538a8d079990a904eab : SUCCESS in 1h 18m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a64c87d163e404189c0b200bf571cf4 : SUCCESS in 56m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65880e994f6840f39b2a022fa73bc2b8 : SUCCESS in 15m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/559df29beb5d404c932f08a394fe1880 : SUCCESS in 5m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/14f8a823e33b4556b6ca30caadc48c3d : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ffcd44e5047448e7a647af35766790c0 : SUCCESS in 7m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a2776013b619496fa565b4030b10bd65 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5041ec014296442ea88a7e4ea6f1ad12 : FAILURE in 45m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0411cd5c59024074afaf3f1f098bd56f : FAILURE in 7m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc21af4f49114645b1a96b834add3d8a : SUCCESS in 1h 23m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85e7fe3874d741e19bd0cf1fa3be9122 : FAILURE in 2h 10m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d98c2d6f64cf4acc949be1241b00c645 : SUCCESS in 1h 52m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bcc0a3bc0b9c4a3cb95e10494a8c705d : SUCCESS in 1h 46m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1cc7cce891e24fec9fb567a278d4aae3 : SUCCESS in 43m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f88133dc61544f9e8a7a3e6eeb1b3d17 : SUCCESS in 1h 13m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/005ec8a6095b495daef2e0af5dcdb53d : SUCCESS in 1h 29m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4b091ff197704a4e94c0db1154646262 : SUCCESS in 39m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/801013e381024f5694088145d95da87f : SUCCESS in 1h 18m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ad0e8ad79d534ae7988d551e9a118a1f : SUCCESS in 1h 38m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/881572ff6f1b41079e027ae8d95bdf07 : SUCCESS in 1h 43m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f20d624ea2ec46bbb0d86a0e539d5bf6 : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":7},{"id":"862415f949ca5e5951bb6e5b61ff4a4bb3445b95","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-20 13:12:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/7/upstream-check/cinder-tempest-plugin-ontap-fcp/98aea0e/ : RETRY_LIMIT in 4m 28s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/a07272b/ : RETRY_LIMIT in 3m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/7/upstream-check/cinder-tempest-plugin-ontap-nfs/465a0ca/ : RETRY_LIMIT in 4m 41s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4aa950c/ : RETRY_LIMIT in 4m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"77bb79be423ec06777a650ef0bfdd849dfe8071d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 15:41:33.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5a913b15c5d741f39060a5bc72c677c6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3e7ced0d360e41169014137b77e3c217 : POST_FAILURE in 23m 15s","accounts_in_message":[],"_revision_number":7},{"id":"eece849399b7357932fa2534bd5a20d821c4a3f4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 16:52:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916465 : FAILURE in 24m 16s","accounts_in_message":[],"_revision_number":7},{"id":"66e54f627e1920e740c312ecbd47ec43e2f0347f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 21:12:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916465 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":7},{"id":"572c7c38f90690a7c1deb914ad3972215ab07cb5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-21 08:21:00.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/7/check/DellEMC_PowerStore_iSCSI/6b07d74/DellEMC_PowerStore_iSCSI/1352 : SUCCESS in 1h 21m 40s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/65/916465/7/check/DellEMC_PowerStore_FC/4ce1222/DellEMC_PowerStore_FC/1535 : SUCCESS in 1h 15m 22s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/65/916465/7/check/DellEMC_PowerStore_NFS/8b9cadc/DellEMC_PowerStore_NFS/1046 : SUCCESS in 1h 10m 15s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/65/916465/7/check/DellEMC_PowerStore_NVMeTCP/1f57c49/DellEMC_PowerStore_NVMeTCP/900 : SUCCESS in 1h 42m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"81c0b4da8dc6461d60980c8f4de0805539c84f1f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-21 14:00:07.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/65/916465/7/fc : SUCCESS in 0h 41m 12s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/65/916465/7/iscsi : SUCCESS in 0h 50m 07s","accounts_in_message":[],"_revision_number":7},{"id":"e6118a61688906656b1578d5bb3b92fdb464bd3f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-21 21:21:15.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/65/916465/7/check/DellEMC_Unity_FC/9442efa/DellEMC_Unity_FC/1915 : FAILURE in 1h 43m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/7/check/DellEMC_Unity_iSCSI/8e3b334/DellEMC_Unity_iSCSI/1943 : FAILURE in 2h 20m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"55559ce85c7d1bb9c1aa69c2e68d4b560ac27931","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-22 19:53:28.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-916465-6/FC/ : UNSTABLE in 15h 1m 36s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-916465-6/iSCSI/ : FAILURE in 45h 10m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"202580ecae3ac3ace4f1767a594c1c71e01a2e81","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-24 10:31:51.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/7/20240524115725/NEC-ISCSI : SUCCESS in 4h 18m 59s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/7/20240524115725/NEC-FC : SUCCESS in 7h 34m 23s","accounts_in_message":[],"_revision_number":7},{"id":"9c12c0c2a0850cdf066b617d7820d49b44b53a9f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-25 05:49:58.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/7/20240525063900/NEC-V-ISCSI : FAILURE in 7h 15m 55s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/65/916465/7/20240525063900/NEC-V-FC : FAILURE in 8h 10m 54s","accounts_in_message":[],"_revision_number":7},{"id":"8f569e7455d04e2458f26490f9324e825fcd515d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-06-10 21:53:13.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"8c3df24eadb2c790ea7f0ce9745f6f06fc1a3ad2","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2024-07-03 15:33:34.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6ef236985b20288a3a955d56b4a2f8aa8f0758fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e289eb63ce4b78f2ad87afeed51d5752787131e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:51.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/595/","accounts_in_message":[],"_revision_number":8},{"id":"14b912d692fd9eb1d04190a4275f1f6018eb5fb4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:53.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916465/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916465/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"864bc47efaaeacb678ac67cdc721f952ccc7322b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 18:14:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b9fda372db24db781f806cddf314512\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e528470db1f249f9a649ba492f506317 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b532a5bd27e406a9fde182b4ac39b9f : SUCCESS in 11m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/801908c6c34c4f38a624a5233fd23253 : SUCCESS in 10m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/87a2a50385b048d59d4a046c21097dcc : SUCCESS in 17m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e56ee93ce749b388554192850d3400 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/63015606570443598b46741612f9db31 : SUCCESS in 40m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d308168d43d14523874138311e3a6c74 : SUCCESS in 1h 21m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4ee8c943e4e46f2b9182ce8e1963a58 : SUCCESS in 28m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/52b0d1f38e8144e2b0a8a9ca7b714957 : SUCCESS in 16m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ffecea3d33fc47d181ceb320ec438520 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/05b2d673fe784586a085708f0d6fa3dd : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff3f343af37847989022ad36c2bec6a7 : SUCCESS in 5m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ab4763e543d04ca5aa847fceb954d157 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/57bd65b14d2b40c0946355045e897ee8 : FAILURE in 35m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4311f80981ca4c28be53653469b88c21 : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f423382fca8946ba97c5125653586148 : SUCCESS in 1h 12m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/77a741a1590f4c79832d468934214242 : FAILURE in 2h 16m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/37ab618850814c81af0a6498d3a1ce9e : SUCCESS in 1h 45m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/eade64c4466e4caf8b6392026c464221 : SUCCESS in 2h 13m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4443ada0d32d4d12b3f8e1f0d5ce6d34 : SUCCESS in 45m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/46c09134174e4b739749d53e2ada255c : SUCCESS in 1h 10m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5a7aff79560d4716b4a20930aa3b2f17 : SUCCESS in 1h 45m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ae40b008a524e1081f364134c480d99 : SUCCESS in 40m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cf25d7cd0cf847079ec30a2548f5b4c4 : SUCCESS in 1h 26m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/20e1c39097664f8d9c1744206f5e5795 : SUCCESS in 1h 30m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6d1afff7045c4d78b0ea950345a5dfad : SUCCESS in 1h 42m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41a188105c5c4f42982af44970351382 : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":8},{"id":"55ac70881448fc26c557448e677af445ddc9a443","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-18 18:19:22.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2024_18_14_14\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"cb9c8e71bbba3c1a63bf480e2722803c8047976c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 18:57:07.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a90073dbc9fd470d9eca0df8749974db\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ce3e89879ff445689cbaeb981cd86c1b : SUCCESS in 38m 41s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ad97f596c1f4991ba2b61ce09425a33 : NODE_FAILURE Node request 200-0000007272 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/253110791cd84f22b489dd81c16c42f3 : FAILURE in 1h 25m 53s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cbbeff085d274b2ca5ad0ad5253074c5 : SUCCESS in 40m 42s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmai... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"32d1a09680a97f4bea070576a909f9dd975c4c72","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 19:23:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/8/upstream-check/cinder-tempest-plugin-ontap-nfs/b8965f7/ : RETRY_LIMIT in 27m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"f231abc408dac3211a40589a8cacda85fcf6813a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 20:56:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8849bcf6338949f18f1194e2e49d6e91\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3861da806053487ba3b9b9f84a8de534 : SUCCESS in 18m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/19ca207402b94b96ae92d616a01ef031 : SUCCESS in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b025f2ff62b025adcfac0848c67bb469d5f63947","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 22:20:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/739d96a/ : RETRY_LIMIT in 23m 57s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/916465/8/upstream-check/cinder-tempest-plugin-ontap-nfs/f600a45/ : RETRY_LIMIT in 17m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"aef588185b355cb09372d92739fe5875daca65af","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-19 00:01:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916465 : FAILURE in 33m 22s","accounts_in_message":[],"_revision_number":8},{"id":"db44ad338fb24cfe77d38a44abcb4775f306106d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-19 04:00:43.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916465 : FAILURE in 25m 20s","accounts_in_message":[],"_revision_number":8},{"id":"2d033b20fe38f30fce36698aa36fdcae58363d69","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-19 05:50:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916465 : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":8},{"id":"966a862c8d42de0c9b64856d2304b3b470796224","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-19 11:13:26.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerStore_iSCSI/bea1dea/DellEMC_PowerStore_iSCSI/1754 : SUCCESS in 1h 22m 02s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerStore_FC/6141ca2/DellEMC_PowerStore_FC/1952 : SUCCESS in 1h 01m 02s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerStore_NFS/138c9ba/DellEMC_PowerStore_NFS/1447 : SUCCESS in 1h 11m 12s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerStore_NVMeTCP/67bd613/DellEMC_PowerStore_NVMeTCP/1307 : SUCCESS in 2h 22m 27s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerStore_NVMeFC/c6d71d5/DellEMC_PowerStore_NVMeFC/271 : SUCCESS in 1h 21m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"8879bab2674eee9e3193dcc542e2a918f6ad21ad","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-19 14:34:12.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/151b06c9ff654899b242df1856a7ea3c : SUCCESS in 1h 20m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/db872f9260584a97ba34365931413438 : SUCCESS in 1h 26m 03s","accounts_in_message":[],"_revision_number":8},{"id":"a2fd5472f801290efa74cff6d50e809412bd89cd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 03:10:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/568/.","accounts_in_message":[],"_revision_number":8},{"id":"0628c313298f3af896f6d350e3ebba6c814374a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 14:57:53.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1156/.","accounts_in_message":[],"_revision_number":8},{"id":"bc6a19923b0a2d3760259fee3bbf394061ea5b12","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-21 11:50:55.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_Unity_FC/4cb8737/DellEMC_Unity_FC/2635 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_Unity_iSCSI/cd238a6/DellEMC_Unity_iSCSI/2644 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"3e5220e58e3858e4fd68da7b4e09fe46c11f3087","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-23 05:21:45.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerFlex_3.6/d6dade3/DellEMC_PowerFlex_3.6/2129 : SUCCESS in 2h 36m 32s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerFlex_4.x/4895ba1/DellEMC_PowerFlex_4.x/1013 : FAILURE in 2h 01m 40s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/7878ec2/DellEMC_PowerFlex_4.x_NVMeTCP/953 : FAILURE in 5h 53m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"8662b21087d6ea1c3c9a92f408fbf7dcb98179cf","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-23 20:40:48.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/65/916465/8/20240923204847/NEC-V-ISCSI : FAILURE in 6h 28m 28s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/65/916465/8/20240923204847/NEC-V-FC : FAILURE in 8h 51m 58s","accounts_in_message":[],"_revision_number":8},{"id":"ea8d180d8d7966da8ef54999a78d1a7965bd1043","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-12 22:57:35.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/916465-8/HPMSA-iSCSI/ : SUCCESS in 3:18:16\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/916465-8/HPMSA-FC/ : FAILURE in 44:23.71","accounts_in_message":[],"_revision_number":8},{"id":"2fc5749514fcb5e96dfe9a06abf74e39b6cf7b9b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-11-15 10:29:32.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fb8828dcbd461f801bb63feb9601d81897485694","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-12-29 00:45:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b0a930ca32529add8b6bc538c82c3b70022451e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 01:11:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96171884fb994b81b1d59a9880c90b00\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64ea0faf21e54a8487b3b68967bfce33 : FAILURE in 23m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c9a8f83f97b495e94a0809a4bd17705 : FAILURE in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a98bdf2337dacf49a7578f6ffb12c6f6f7333d12","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-12-29 01:54:07.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerMAX_FC/7155d45/DellEMC_PowerMAX_FC/469 : FAILURE in 11s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/65/916465/8/check/DellEMC_PowerMAX_iSCSI/e9fe42f/DellEMC_PowerMAX_iSCSI/481 : FAILURE in 41m 41s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"9f4748b7c495780a1ff834be29a484311e1f2a6f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-29 02:07:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2a667a6871a24d7fa5041f211d5c851d : FAILURE in 1h 17m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/abd5ff4e18f74424ae2c729df57ee388 : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":8},{"id":"8ef3efdd95e9f40144d77a0c1f74903afa03441f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 03:26:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9a2f86f67b84a3f96df4e3c845b1eea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aada56d65f2f44749166ab00c6abecfd : FAILURE in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/430af6511b6a4ff1929ba39bd2f429f8 : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4c20bca33a545ba879c96bf3e8a096c : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e89141bc027d4283ab294caf56f93a0f : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/71be4319937b457db9a908f4f6b7b18a : SUCCESS in 39m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6c99acdd38446f798893934979b1f02 : SUCCESS in 50m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b26bbc5ff924f538aceea8b85b760b0 : SUCCESS in 1h 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7493659058494edfab2ff393f7fd5515 : SUCCESS in 52m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bdf9fc83437f4ff69fbf32e39264b295 : SUCCESS in 18m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0fbd8ce8034d420a803eb82931269c44 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7632d7e429f41c1bbef0b1a1e6ea92e : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25b0219d026e4229b228a1e34ef90ec3 : SUCCESS in 6m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c0a1c0b83faf404a8867d4cb73135c72 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/12715372352d48d694a6375a9f3a8f55 : FAILURE in 19m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/32d9cc97d29d447d887e65c8ee37a104 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/20e1657b5de54cd0a1077fc4dddc790a : SUCCESS in 2h 07m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ca623c657eea4aa8a2850fea165f0caa : SUCCESS in 2h 06m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4e6b4ab3790f49fc90567697ecac544b : SUCCESS in 1h 59m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8f07a43341fa4a41a414b36ca4528f9a : SUCCESS in 2h 00m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a7b9a553532a4ea68e8680ef74bcab66 : SUCCESS in 40m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7ef35f3b482f41e99c000a35067886f2 : SUCCESS in 49m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0aa8eec105914cea8af27b0e0b419f42 : SUCCESS in 1h 43m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c9a7fb53a7b2446e8dc3432bbce06668 : SUCCESS in 44m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8ea13b411e534a4bb7ea8ec3a7db3517 : SUCCESS in 1h 00m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/33ab130ce87a4500b9fa6cb4f78aa359 : SUCCESS in 1h 33m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/12771d0f0e3144538507208dd9121290 : SUCCESS in 1h 17m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24a023017d1d4c9ba07e43df7848cdd1 : SUCCESS in 1h 11m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc3748f347a54004bd3540af32228a68 : SUCCESS in 42m 24s","accounts_in_message":[],"_revision_number":8},{"id":"366f5efb24f1cb2a08609501141182c4d3554dc3","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-12-29 12:31:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5324f119ebe845b93a2103e5f13358801ecdb76f","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-12-29 17:13:58.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e086b2e97d4ffb48314aeed79ef3891cb3a597ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 17:36:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d88892817394b619ff4e10cb2ec6c2c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e85c26a131794d33a0c62618e4fe3aa8 : SUCCESS in 20m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f720637822341d986bf416f713e4ddc : SUCCESS in 19m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"70cccaad7cd3ccee05657d1591fcf0c6c909e115","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-29 18:43:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ca59e73f78b8481b9eb14d4bd7a9ee47 : FAILURE in 1h 14m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f1700c1175b2484bbe4aa4523147afc0 : FAILURE in 1h 26m 34s","accounts_in_message":[],"_revision_number":8},{"id":"90e61d1d3f97160ade7ea06a408df85d0266a2f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 20:01:52.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cce0e7fc11f8480f8f86a833d6f1b98a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/942b96f48169495d8c13bc5fbf068198 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/099e5cb9ad8948f19f0824cd00a76f9b : SUCCESS in 12m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0bea6b2ade8141f289b2a3b65e9fc0af : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3f6637d3dd4fe889484e865cb5fbaf : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/63331545bdfb4214ba12630e959c179c : SUCCESS in 46m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c44525aea8f5405aac0a851d671eeb84 : SUCCESS in 1h 04m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7cedd892c64a4690afafd5c8c4c741b2 : SUCCESS in 1h 34m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51b2baea10ab4c599efdf7dea5c45047 : SUCCESS in 1h 03m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c7657da16e84881a8582cfc88de2a6d : TIMED_OUT in 41m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/edd0ff4d94a64f88a104f8455b23ad86 : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f237a48fc3949f19b2f63e7f21edd9a : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58513f54239941529f8dafc8d67bbabf : SUCCESS in 5m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b1134c9984a24c70bb45c6ac56513d10 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5ce710eca4014f138938ff771cedcc84 : FAILURE in 38m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/426d4f500e8545108a7e9964174029b0 : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a2039d69d8d484db3f436f296c12f2e : FAILURE in 1h 26m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/29ca42d544d54457bca0fecd5415fb15 : FAILURE in 2h 19m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/36f37b2136ce4d8aa50708cac1203c34 : SUCCESS in 1h 25m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2b939a8a755e467a947c7b51e3c4d6c9 : SUCCESS in 1h 34m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9f28a2248b57400aa87267f773b55e91 : SUCCESS in 26m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ca419e4593d6467fb393a247ce1f0eea : SUCCESS in 1h 17m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/569f3eda878b4a199e4bd4b9205e577b : SUCCESS in 1h 09m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/37c15fd9a28d476dbbbc150b36c343dd : SUCCESS in 43m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/35acd7b10a89446cb8bf1bb5f5631101 : SUCCESS in 1h 15m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2191de0a2b5d44558b1fece92ea401d1 : SUCCESS in 1h 09m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e762f983e61a4bea8701b7c62068cb4d : SUCCESS in 1h 14m 04s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ade00a369fc049c481f8df07bf549042 : SUCCESS in 1h 44m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/481b615161814c09a68e90c7ba866209 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":8},{"id":"e1fd6a8e0c3a43cfb1163663c43b69131dc11df0","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-12-30 15:10:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0f57849ba53092957ec19c7993a409b804110dc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 16:09:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a5bbaea7b314ed5b2d920941e8a182f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/548925c119004e3baab679b333979366 : SUCCESS in 24m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/feb1fa86ad644eec8801afb571881360 : SUCCESS in 19m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"21299147bf9d5c79fdc5399911445290ba105604","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-30 16:33:11.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c744d21b5a804852a284586f7680324e : FAILURE in 1h 14m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/42346f6bc16d417ab1781f96863c962d : SUCCESS in 1h 19m 04s","accounts_in_message":[],"_revision_number":8},{"id":"326c697f6e719e2638b60495da704a19efaa8272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 17:37:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5e4743b392c44208cde16af00c74032\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee259629eb448e7843a6d07df948e9b : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9a3a9262e984d7ea2c58cb4dd37c599 : SUCCESS in 14m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12a29afd9876437287075b548aaae0c2 : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef5587a49cd4598a44ed3c14c4f2de9 : SUCCESS in 17m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c185eb7911e44263921f30d46015bbb8 : SUCCESS in 1h 06m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f51281820f54827acf7d29846ddd24d : SUCCESS in 55m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ab4aa3f6cb04799bed4af8b0530fb34 : SUCCESS in 1h 39m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e112c0d64cae4f49a33d7efaaf0ebd9c : SUCCESS in 1h 05m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5f2b71476da740cab924ee226eb29022 : FAILURE in 24m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/94753ff58593423caea01ace6413ffbe : SUCCESS in 7m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/52de0f8b4ad14fbaa72795393fea7bbe : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b00db53f5b094adc995e07c58278102d : SUCCESS in 7m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/244567a219e8415fb7027ef24efbc227 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/98c54f11b2774364b72147029d619e27 : FAILURE in 27m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d05ed163174e430f91af9f947bec1710 : SUCCESS in 7m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd120960f52044448e1059190e9b8610 : SUCCESS in 1h 46m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d7236396d43a4a02aa6b15567ae0491f : SUCCESS in 2h 25m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/04771b5bbf1c4507b7983634162590aa : SUCCESS in 1h 45m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/63648a8872284a9f84f4d2585809138a : FAILURE in 1h 54m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/27c4b25dd9b44aed92c8f36955dbef71 : SUCCESS in 30m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4759e9aef1654f9ab1d07c97bc5805b6 : SUCCESS in 1h 08m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/26a3fbfe867c4be2be1c4dfc77fac9bb : SUCCESS in 1h 29m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3c309c00b7b74a1d98e2b6d27c804e25 : SUCCESS in 57m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b76e8e5b9a8e4de7900aa11f06239443 : SUCCESS in 1h 21m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1392e22cbbf248a59ce1217c961d230e : SUCCESS in 1h 55m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c608e1b38c34dbcbc0cc79dc5fa0eb3 : SUCCESS in 1h 44m 32s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1edab7c927c487d83631e9e61d16aa6 : SUCCESS in 1h 06m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9d9f5a80df34b62b16a0dcd018b2763 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":8},{"id":"fb1292272b31ac0d344b642f2931965b8f8a0d2f","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-02-14 16:25:59.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"6693a8120efcf27d4f0b77ee3e63dc3fbdf58cb5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-21 06:30:02.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/65/916465/8/20250221140236/NEC-ISCSI : FAILURE in 52m 38s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/65/916465/8/20250221140236/NEC-FC : FAILURE in 1h 27m 23s","accounts_in_message":[],"_revision_number":8},{"id":"5000e4595431be7d0ccf99d60dc3acdf9a5fb10e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-27 14:04:07.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"13fb79adfc3b0b9e46955cbd30fc61a63b8c6830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"8167b76b19c297aa4ed54f11a40a3516d7cf8335","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"e43e9b93697538e9131da8cdf0c7a3e435af101b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 12:53:59.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"4211ac749086247b2af630e53eded0dae3c10ec8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:54:36.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"83456dd2d86e11a23c3fb547a02ab8a09b201ff2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 12:54:38.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"009d100396706d36178b576549cb3ebd33ac3d5d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:16:47.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"52fb6b0390497156ccbd6dc9ae99d76a2faf2ef9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 13:56:03.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1238dff20d6fcce3e3c74b446b547052f297d8b5 .","accounts_in_message":[],"_revision_number":9},{"id":"667e7b552f0381e0b2b8920bde66db0803d09df6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 14:53:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84648d93255d493297327cdbbf041ded\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/467cf2ae4f614b7a97549c4d8854e31b : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb80b5353fbd4c4cb9c03d7602fb6d1a : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9bae0287d85547638a695500339ed9ea : SUCCESS in 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0a20a0d67f3b60dd604b4d6688eaf90a7f557b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 15:13:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8c279e1b81a462cb7ff3498b77f53af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2bf0ccf4ba4843a008f7e8a1731cdf : SUCCESS in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/facb68334f4049178cbc5b52f610707e : SUCCESS in 11m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe8b6dcd01dc4ff4bcec0c25caaed592 : SUCCESS in 15m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7623319c17e40b0b8b7f4b82a548876 : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/0ec076d28222469b87c76c569beca1c3 : SUCCESS in 1h 11m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b9cf1f411a14ae7bb12af83ae781f9b : SUCCESS in 1h 00m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eee48c48831f4ebeb2a32fbbf6c13e15 : SUCCESS in 1h 34m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a00fb25b99b49d2a028e4ac2a82e625 : SUCCESS in 1h 03m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5a7cd29e2d694def89260fc8c3af140d : SUCCESS in 25m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/70d01268b722429d84374742c471c0a4 : SUCCESS in 5m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a46d33fa78a640a08e6e9476af4cacf7 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1ce47e8a1ba496c8be4bd6fc8d6cfd8 : SUCCESS in 9m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6531146dd9104dd2a77ba25c2654d076 : SUCCESS in 9m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5834a7dea3984973a28bd09cb1bed6f9 : FAILURE in 36m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/23556718a04e4ee0b15242ffacd91a67 : SUCCESS in 8m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d71c6de63dee485dbd1d808dde9b92fe : FAILURE in 1h 05m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ae4ea475af60446d81afa5c14d679dc9 : FAILURE in 1h 31m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f225fa29c42c494eb778ee29da01b73a : SUCCESS in 1h 54m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a8182a4af8824c15aa83bda4c78f4382 : SUCCESS in 1h 50m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/227a6938a2d94a08a6973a2bf826eabb : SUCCESS in 18m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/40d86fc5ee574b5bb226a1535d95950a : FAILURE in 40m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f9c6bdc06a2444b8a1b89bc99c67a1fd : SUCCESS in 2h 10m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/757fe044895c4db986d596b110e42a24 : SUCCESS in 1h 21m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3a87b2db78584cc79fe71c847fdbc7d6 : SUCCESS in 2h 01m 29s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ab1f6cbfa90542d5896ac8d7cbc4f5ff : SUCCESS in 2h 08m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f44c3c3d69d84ff6b670b4fe7d5cb6c0 : SUCCESS in 1h 50m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e62c42cd1ce14453be01f455b72de68a : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":10},{"id":"039bf75d43f5b03a15af5b52d642f8383db46dc6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 16:50:43.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1238dff20d6fcce3e3c74b446b547052f297d8b5 .","accounts_in_message":[],"_revision_number":10},{"id":"f4c17e829215deca1079978c3313c8f5fa59838b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-25 02:27:14.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aacbe3982ae541c6a115bbfbddc8dbc2 : SUCCESS in 44m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b2f1dfc4288d4c8fa0816b21a9180141 : FAILURE in 59m 39s","accounts_in_message":[],"_revision_number":10},{"id":"ecee8d72e75c03100afe900f0dedce38b6e3e006","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 03:04:19.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2259/.","accounts_in_message":[],"_revision_number":9},{"id":"7c4b7a84884aaacb135d7d62c6e043384bd77d36","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-25 03:10:18.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916465 : FAILURE in 1h 12m 01s","accounts_in_message":[],"_revision_number":10},{"id":"1f33d183a556351296ff7385d56c8c308b3145b9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 03:28:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2145/.","accounts_in_message":[],"_revision_number":9},{"id":"b26e29bc57e6193cc0e0c10928c8ab9823785798","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 18:28:21.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2277/.","accounts_in_message":[],"_revision_number":10},{"id":"3f43f5fe277b05618f622c4440bfff9578bfaeb8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 19:15:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2163/.","accounts_in_message":[],"_revision_number":10},{"id":"d634b6170c14b61c199267380a81d2579d24960e","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 15:09:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"761a55d70799b5ff1731b167d916e6e4eeca7c17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 16:32:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd0572f06e9d4b318cd8546f8e572b03\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c63f677d71294c25ad7b4fd75b3d2542 : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/34e95321bedb413ea3b63767602176de : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3fb55a31827d4ddcbb9a282d508dc017 : SUCCESS in 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9cecf5c8ee69c09f4a0674963e7918d24e1ff560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:06:56.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aea771f557d14a13a163883b77c7aee4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e30ec8108a04c8da901b6c381ab3d5b : SUCCESS in 7m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26ecf8ba5dde44de98b62c403240b89c : SUCCESS in 13m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5892fc192ab94cf6b72108cf0bb62496 : SUCCESS in 16m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde22cfb80884d438843e4e450943322 : SUCCESS in 7m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/9010dc17a948421d8a2d1d8e086c5921 : SUCCESS in 1h 00m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3fb2ca9652864f5391c8d024e7355d81 : SUCCESS in 1h 06m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff3f504ab2e3402abc16533c5000bfed : SUCCESS in 1h 31m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7055f499028c4a208118c80043ef376b : SUCCESS in 1h 05m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c882b7fe5b44d7d84b8a66594a48d76 : SUCCESS in 23m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd7867da2830475d9a0ed08ca23b7970 : SUCCESS in 5m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1358251185544e194097b888b07f738 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6df67544a5264b17b000842f887e417d : SUCCESS in 7m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a6b3a00739254243858babaf9573a480 : SUCCESS in 9m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/200c1a893cca445da8da41a38c7d9fd9 : FAILURE in 40m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/df064fa547ad4ef7a0346a365f836748 : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8c6741e7de047db943dac04f2f83678 : SUCCESS in 1h 51m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0adb41199a0e444a835a67830231fe61 : FAILURE in 2h 51m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cd204a140a7d492da5b2d52e8c2a9c5f : SUCCESS in 2h 01m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/91f117382d314dc2b42979563fe020f5 : SUCCESS in 1h 56m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ade217212b9b40cc9a51298b5443ef2c : SUCCESS in 37m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ac19ba6a2bdd4ae2895863af19694519 : SUCCESS in 1h 08m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7d19d95a4e294126878c177c0ca953b4 : SUCCESS in 1h 48m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5916274abe054e1b8013889df2947977 : SUCCESS in 33m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e8b1e829317d40fa94eb8dc5614ab3fb : FAILURE in 1h 04m 40s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7b16dae703fb4fd1811611f78a8e4f5b : SUCCESS in 1h 15m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f54382d275e548c6bb02d007d1c2e70d : SUCCESS in 1h 52m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc73613179cf4a95a6e2920df5b51b2f : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":10},{"id":"d2b4877812182bddf0e7188336f19f8f557596dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-30 19:57:38.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c4a8851eb39b41d684bdba6fb86c5943 : SUCCESS in 49m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/22d7297fb09a4030a57445d27427eee4 : FAILURE in 1h 05m 34s","accounts_in_message":[],"_revision_number":10},{"id":"ed4d24fa1dc89e75a837c6bf7d80e97fe1372364","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 10:22:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b3c11e99010478e9a9f6a434674e3e41f1c23f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 11:51:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09378a5388874ae8a3117b12d2ab671f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30462481d6454e2486c45377817492f8 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1420181d3d6d443cb4534e27af00d581 : SUCCESS in 20m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6400bb56290f435faf0d7bf44f0dd41e : SUCCESS in 30m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5cfc771e6906dec7ab34a9aa66e84e8f208ed8d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 13:07:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/524a02a3a5914a98abbd5b2ec38ee2b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cff802fcd9f64f51913835b556f9c719 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cebffa6dbadb4d3f97fa4c85a7755966 : SUCCESS in 9m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34bc971098d749d7835bcf4151e08fd6 : SUCCESS in 13m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a640049730c49da944aba3f691de27d : SUCCESS in 16m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/33f40c187f394a329aa9986799e99309 : SUCCESS in 1h 01m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94e5eed860824c0c91a9fdea2947dd3e : SUCCESS in 1h 02m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c381ee6bd7344838a1c05a8a0eee3a3c : SUCCESS in 1h 45m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d292ccee30d94417b64371c12f0d6666 : SUCCESS in 59m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5345b105d049409ca11a9a205ae0e8e7 : FAILURE in 25m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9931e13b86ec4f24bf4f8fcdf5489b5e : SUCCESS in 6m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de2fe58fceb74fb8875527b00372d471 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce252fac5066433c985519ca9b0ec71a : SUCCESS in 6m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/716ea90fcef84980aa30de0a35bea209 : SUCCESS in 8m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/81abaa8a2d3d4872b08ca8e4c3ebb231 : FAILURE in 46m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6d1a8a9a6188424fbf1fff3111a494cb : SUCCESS in 3m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed4d2e56bff74b2db5bc4a9bd511a6ec : SUCCESS in 1h 50m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3938a713a7724062975c437527d40678 : FAILURE in 1h 39m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6dbbfd1f345646a9ae69be566886cbcc : SUCCESS in 1h 39m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/354c6774bd9845a699da8e17dd5fc529 : SUCCESS in 1h 55m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c75b2451695a42659bb45e5a7a8cc949 : SUCCESS in 34m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ab04aa7033b482fb4ae800d2ed39393 : SUCCESS in 1h 15m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2c8efebf8ef845c3972e43dd00226f14 : SUCCESS in 1h 10m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/451de495fa6b4caebe8c3efdabc5696e : SUCCESS in 41m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a5f47e1d7a240738e68c2580fd2a7e5 : SUCCESS in 1h 52m 12s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4a6e4d9ea3d84fc4ad8883d2b8844b54 : POST_FAILURE in 2h 37m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/942b5ef289c840f4b7a6eb6e303fe06d : SUCCESS in 1h 33m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97cf8d5842b446159740a1f2a4915fce : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":10},{"id":"bedd5fcc3ea0d1a979e008a1d56fb45e6ef6824f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-31 13:25:49.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_10_2025_13_19_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"b347185762439684c1ff3da8fdf5f209ad91a874","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-31 13:58:50.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"f939a8d61b28cc4144f6c5f7e31635f59fd6556b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-31 22:30:16.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9b66863a3d644c1e9fe091abfe9a42c0 : SUCCESS in 1h 39m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/202572b01654481cab85fcd7c282fc4d : FAILURE in 1h 47m 12s","accounts_in_message":[],"_revision_number":10},{"id":"e1357bbc2f046dca034e8ad41eb576a936736a5c","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-11-07 02:28:42.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916464 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"7817e64588eae54a414b4f7e7533979bb13dfaff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 02:29:05.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a3befd91b85125545a1792769d55de602bfd82be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:35:53.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6887b97af8fd44cda75ff89a30104ffe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dcdaac3e4234d3383928f6f05ff9b5e : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df79aee7e14f4effb9661651f1a87960 : SUCCESS in 11m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2404fa47f61f40439dba2a245ba19e6e : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6db0fe889d4a97a9410ff078935459 : SUCCESS in 14m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/0e39deefb4e64f7e93d970ad84a1358b : SUCCESS in 57m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad402ff6fd8442379abb30598a558314 : SUCCESS in 50m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aab70888734b4362b6db59f8a64d0515 : SUCCESS in 1h 37m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2e5ade17c364c84b6aa3810df79eab2 : SUCCESS in 58m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7deedb4b7962406b8e726245ef072e26 : SUCCESS in 1h 14m 04s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7206b2fdb8f64063b4c0c0ed7c87af30 : SUCCESS in 1h 59m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7395ab0f623431e9b9b0ca9bf18c160 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":10},{"id":"b6b55140bec30801cadfbef0672d41f613bf4b7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:35:53.000000000","message":"Change has been successfully merged\n\n8 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":10},{"id":"7cf818c5995fefda7315e46e091b428f98a1fad1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:36:53.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f28013f23534356b76321f2cfbbcf7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d18259acfb864156ba3ccacdd2450102 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"91a5db818b7603c34f822a856e80fc82df03a35c","revisions":{"c75e935557c45c8af76d286b4616f0432daeb903":{"kind":"REWORK","_number":1,"created":"2024-04-19 14:51:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/1"}}},"commit":{"parents":[{"commit":"85b3b7cef63f0b35ea4fdf0f6d2c12f5562a5636","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85b3b7cef63f0b35ea4fdf0f6d2c12f5562a5636"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 14:47:56.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c75e935557c45c8af76d286b4616f0432daeb903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c75e935557c45c8af76d286b4616f0432daeb903"}]},"branch":"refs/heads/master"},"6cb93932a18bd1789fe5f7cc4d7ceca066b43023":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-19 17:27:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/2"}}},"commit":{"parents":[{"commit":"db348a3efeb27a79bfdb80be504e03650a160068","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db348a3efeb27a79bfdb80be504e03650a160068"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 17:27:38.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cb93932a18bd1789fe5f7cc4d7ceca066b43023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cb93932a18bd1789fe5f7cc4d7ceca066b43023"}]},"branch":"refs/heads/master"},"db09b0c40c941584f2d81c6deb75c0396f864763":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/3"}}},"commit":{"parents":[{"commit":"2bc5d5f69070bb71849969080fb7743d46e151d5","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bc5d5f69070bb71849969080fb7743d46e151d5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:51:11.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db09b0c40c941584f2d81c6deb75c0396f864763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db09b0c40c941584f2d81c6deb75c0396f864763"}]},"branch":"refs/heads/master"},"80328ba3fe5902c3716cf1c4680d564950c0f0cc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-25 13:02:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/4"}}},"commit":{"parents":[{"commit":"6afdd9bbb018b8042f35e1a104f6f3537adc307f","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6afdd9bbb018b8042f35e1a104f6f3537adc307f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-25 12:43:38.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80328ba3fe5902c3716cf1c4680d564950c0f0cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80328ba3fe5902c3716cf1c4680d564950c0f0cc"}]},"branch":"refs/heads/master"},"a1d32b4a55ba7745ed2397a7ac87eca4b14ca86d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/5"}}},"commit":{"parents":[{"commit":"5ba63fda049aa1f36d777998f1bad51fd9e4556f","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ba63fda049aa1f36d777998f1bad51fd9e4556f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1d32b4a55ba7745ed2397a7ac87eca4b14ca86d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1d32b4a55ba7745ed2397a7ac87eca4b14ca86d"}]},"branch":"refs/heads/master"},"30521f7dafe600fb7ad6297be68189c4c10bfc3b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/6"}}},"commit":{"parents":[{"commit":"809420d1ac6026222d56b49f28e5e9b670b20aec","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/809420d1ac6026222d56b49f28e5e9b670b20aec"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:00:19.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30521f7dafe600fb7ad6297be68189c4c10bfc3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30521f7dafe600fb7ad6297be68189c4c10bfc3b"}]},"branch":"refs/heads/master"},"b0fef320b60a83d9abbeab57a643e6ea8e3a94c3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/7"}}},"commit":{"parents":[{"commit":"3c8acd3ff967e2487236e95b08d72b688327028f","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c8acd3ff967e2487236e95b08d72b688327028f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b0fef320b60a83d9abbeab57a643e6ea8e3a94c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b0fef320b60a83d9abbeab57a643e6ea8e3a94c3"}]},"branch":"refs/heads/master"},"5bd4bc8b2939ab0299134644f4d7f2f72ece1f7a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/8"}}},"commit":{"parents":[{"commit":"5c4f29320389c00ba39dfc25067a4b56ec6eec68","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c4f29320389c00ba39dfc25067a4b56ec6eec68"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 13:01:44.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bd4bc8b2939ab0299134644f4d7f2f72ece1f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bd4bc8b2939ab0299134644f4d7f2f72ece1f7a"}]},"branch":"refs/heads/master"},"ad45587ee63f62505eef1e756717cb065fda0533":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/9"}}},"commit":{"parents":[{"commit":"6ecc53f8df97c51c791c59b974c28b8392a4a515","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ecc53f8df97c51c791c59b974c28b8392a4a515"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:47:28.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad45587ee63f62505eef1e756717cb065fda0533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad45587ee63f62505eef1e756717cb065fda0533"}]},"branch":"refs/heads/master"},"91a5db818b7603c34f822a856e80fc82df03a35c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/916465/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/916465/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/916465/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/916465/10"}}},"commit":{"parents":[{"commit":"3d296ed3d6146cb88b59dacb54df46ac748d580b","subject":"api: Remove unnecessary router split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d296ed3d6146cb88b59dacb54df46ac748d580b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-02-21 11:03:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:53:11.000000000","tz":60},"subject":"api: Add missing functools.wraps","message":"api: Add missing functools.wraps\n\nTo preserve function signature.\n\nChange-Id: I1238dff20d6fcce3e3c74b446b547052f297d8b5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91a5db818b7603c34f822a856e80fc82df03a35c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91a5db818b7603c34f822a856e80fc82df03a35c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
