)]}'
{"id":"openstack%2Fcinder~916467","triplet_id":"openstack%2Fcinder~master~I174c3bae39625b74e4b416e3559eb79ad8d34948","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-12 17:44:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I174c3bae39625b74e4b416e3559eb79ad8d34948","subject":"api: Keep track of extension controllers","status":"MERGED","created":"2024-04-19 14:51:42.000000000","updated":"2025-11-12 17:45:41.000000000","submitted":"2025-11-12 17:44:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"916467-openapi","meta_rev_id":"e2ecabdb84acb0f48abdbee0ee80c8e2d59e7742","_number":916467,"virtual_id_number":916467,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-12 17:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-11-12 15:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-11-11 17:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-11-12 15:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 15:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 19:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-20 05:19:53.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-20 06:26:38.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-20 10:13:35.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-04-21 01:12:00.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-21 07:15:10.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-23 09:05:38.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-24 08:56:18.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 08:06:30.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 17:04:10.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-04-26 19:13:11.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-26 19:28:46.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-27 00:09:55.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-05 12:13:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-11 02:30:26.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-25 22:25:43.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-09-18 14:15:45.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 20:34:04.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-19 01:12:40.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-19 17:24:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-11 18:33:22.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-10-24 14:55:17.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-30 15:08:30.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2025-11-11 17:25:41.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-11-12 15:27:40.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-11-12 15:27:57.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"30bf9eb15f9c79a41b8449231b18bf88fa224a40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 14:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff2e775d3610eb95c2f8a392648fab435bdd28f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:29:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6579f0f536764552bafcc8dfb6fd8112\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3d4f940afae4850a00c6cd750b357ca : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8dbb768ae7da4945a42a2abc3ec1d1f8 : SUCCESS in 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aba70495be393b9168edaad17e621f73c4254c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 17:27:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0ea3edd34b2df4961ab24ead399f8deb2a59bc90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:46:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f9c8ea6bb954fbc92dc16c07b17cf72\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23a438733cc5492fa883949008d60de3 : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40020d3bafa944c7bdb829a36f84d501 : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"307d6f15256b11a4aae510d2e3e7eaa9e63e0f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 19:47:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52e99cfc32e043a99b083522388ca71b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04fbc57cbe9d462cb5eb1b4e1823ef6b : FAILURE in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/615a2c9f4e5546048535752884d621fa : SUCCESS in 10m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c3c3ead48b945babb338e95847199e4 : SUCCESS in 13m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cfefe9d122694205886e01b095b41eeb : FAILURE in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be38f05d826b489a8453cc2488037e06 : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/9aa062f8a80349298763b4b3f49b0dd4 : SUCCESS in 43m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72bc70de9a6e4e239f3f988f309630ad : SUCCESS in 1h 20m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b7d773cf5264908ab1c40d01d833bc6 : SUCCESS in 1h 00m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3c28f05ea4754fe49ffc26a32ae81a08 : SUCCESS in 19m 13s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/372a807ab85f4dd795a7fabe2c3383e3 : SUCCESS in 13m 17s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7e36e6e6659b42e1820ded20751cbb6d : SUCCESS in 5m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/96afaa5b7fa04a6bbe15c927986b7d32 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29a97dfa36704dc6977a31c71258a6b0 : FAILURE in 6m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3409c1f0b1804998866cb894033bd24f : FAILURE in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f8977c18756e4aeeaf5240ec98e823cb : SUCCESS in 45m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b13517c9a93247db963e1cd9e675ecea : SUCCESS in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8e03b0ce1d24d378b9216dfbb64ad76 : SUCCESS in 1h 54m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/29c12f13d2a347839efcae41fdecbcea : FAILURE in 2h 04m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1aa25dcc317e41ba8118bdad471b0e23 : SUCCESS in 1h 26m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1d54915ea23b4aefa351983cd21543be : SUCCESS in 1h 36m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e874d915510c4862bde859c7ab2203f4 : SUCCESS in 34m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/651f829492fd4937bf435e86e262f0bc : SUCCESS in 52m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e6764b9bf8774dce8b2b0da8e9d49e62 : SUCCESS in 1h 07m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/286cdf2d5e79465aa1f13f07db3d8704 : SUCCESS in 39m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f936c01f2c6d4cb98849b999afffe9d6 : SUCCESS in 1h 30m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3529eced1a5c4c099e5cbb7f36e4a320 : SUCCESS in 1h 46m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39bc4f86b955444db62efc0e343836f8 : SUCCESS in 1h 11m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bad3c4da40549189981aa4ca319c0c0 : SUCCESS in 52m 30s\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/api/contrib/test_admin_actions.py\n  Comments left for invalid file cinder/tests/unit/api/v3/test_snapshot_metadata.py\n  Comments left for invalid file cinder/tests/unit/api/v3/test_volumes.py","accounts_in_message":[],"_revision_number":2},{"id":"abe5a5034ae98990d7206f71a248994d304a4af4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-20 05:19:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916467 : FAILURE in 36m 19s","accounts_in_message":[],"_revision_number":2},{"id":"70790134e239f056079eaa4cabfe2c3832ef8b3a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-20 06:26:38.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/03c194629dbc441594ebab5289161ccb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f83a8a2c53e049058570380ae849d061 : FAILURE in 1h 31m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2f5e479b8e4907d0e1bac76f819246f2b53a9cc3","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-20 10:13:35.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/916467/2/fc : SUCCESS in 0h 39m 55s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/916467/2/iscsi : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":2},{"id":"7260e3017c703ca61087bceb9ced57d24d35ba4b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-21 01:12:00.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/916467/2/check/DellEMC_Unity_FC/39ccf0f/DellEMC_Unity_FC/1720 : FAILURE in 1h 37m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/2/check/DellEMC_Unity_iSCSI/42f8b0a/DellEMC_Unity_iSCSI/1750 : FAILURE in 1h 47m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"0a7106a8890500640d9cab1aa8c6c96c3f53a6a9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-21 07:15:10.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-916467-1/FC/ : FAILURE in 27h 51m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-916467-1/iSCSI/ : FAILURE in 27h 15m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"6fcac1f75114a489bb044d1abf8a766b1537141c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-21 10:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/916467/1/fc : SUCCESS in 0h 43m 53s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/916467/1/iscsi : SUCCESS in 0h 49m 35s","accounts_in_message":[],"_revision_number":1},{"id":"162b8cc98f59147ca9b39741e6b0890f16842e82","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-21 23:27:55.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-916467-2/FC/ : FAILURE in 29h 25m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-916467-2/iSCSI/ : FAILURE in 27h 59m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"7ef2f24778801f516d88d181b5ca588bc0299c41","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-23 09:05:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/261/.","accounts_in_message":[],"_revision_number":2},{"id":"7df9eea28331f9857bf77e0c0575a61f320b8261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ab722ad11edc1ad51a2b8d58127aa1565a393d9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:19:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa0aba928324304b1dfed07430b60bd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83fed4994d0b40aa984d400fed9a79b5 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b510cbf595e43a5a596e31b9b099f8e : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f21032018709533b7695b12ee442bdcb7199b596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:23:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc3794fdf70444ac8c1b30758db6b514\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ce9a6d5b6e44939d44389325deff6b : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e6ffaa5594c433cb204e44a56fbd51f : SUCCESS in 11m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94d66ad2c76045d7ade3ef737335aef2 : SUCCESS in 10m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3639899f1a704f758691c6f579633ec1 : FAILURE in 9m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/575a19a5db4e4ace94907b86548b48f6 : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a6e1c2bacc054d389890aa6c07573b3b : SUCCESS in 1h 01m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec5bec7ea4d449c894e4f772a05cd856 : SUCCESS in 1h 17m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70ec49e856c24e67a6e2963577832824 : SUCCESS in 1h 08m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8fb5280cb4844007bdf164d241bedde6 : SUCCESS in 15m 42s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6050a1dca9954cb0a78be913b63d00f8 : SUCCESS in 12m 49s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aabf39f435474f2fb6cd47b15b5f1915 : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb85aee302704753b77781d7e609f23c : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ddc3f50d9454c738df7aaee6dc80e6b : FAILURE in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f86eb2c5e43340d28bb3ae6977a81ee9 : FAILURE in 7m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/538df5e8b6184279829accd8da4dd026 : SUCCESS in 43m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b833b1a24ec4b91969dfff3c4e9682a : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c96b5f8a59034262bfa613899ce03039 : SUCCESS in 2h 10m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/22eea8cbd7c6440ba0015db10d0eb41e : SUCCESS in 2h 17m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/663dc1bde36240a8a77c0e3beccdafba : SUCCESS in 1h 40m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2f0989c50dc146f9a3dbb11f41b4ec09 : SUCCESS in 1h 51m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c196f0e5a4ac4bde9145b1b86e6b3de8 : SUCCESS in 40m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5273e7eb10a74820a8162e381af8271e : SUCCESS in 1h 25m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/90c8c1e99ee34ab2b7dda24c8dc85433 : SUCCESS in 1h 22m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fa10989f2caa45c48d3aa5122f990e0b : SUCCESS in 42m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ce9932a698d9406682843c79628e8445 : SUCCESS in 1h 24m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/914b86b9d46743beab3dd00603206bd0 : SUCCESS in 1h 18m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de09868f5c664775999d316ab0f31589 : SUCCESS in 1h 42m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea7ebaa0ce774d2c9c36c54afda0d757 : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":3},{"id":"5c331f92c0659c584dca70c05d13a28b8e9eacc0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 17:27:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916467 : FAILURE in 12m 49s","accounts_in_message":[],"_revision_number":3},{"id":"f977225e5d891bf0988e1b7309fb3f7e8645fb47","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 17:30:21.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2b265dceaee04000951f031f7606521c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bd918560c39c4ff4b53fed5f7230fbb6 : FAILURE in 38m 09s","accounts_in_message":[],"_revision_number":3},{"id":"d2cfd4fc5d46d937b230d555b0b0d485449669db","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-24 08:56:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/67/916467/3/check/fujitsu-eternus-dx-iscsi/22db9ff : SUCCESS in 3h 37m 16s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/67/916467/3/check/fujitsu-eternus-dx-fc/6ed38af : FAILURE in 3h 51m 23s","accounts_in_message":[],"_revision_number":3},{"id":"64b2bb951eab4afb94fb6727e1fb227f16ab1cb6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-24 17:32:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/916467/3/check/DellEMC_Unity_FC/4901d10/DellEMC_Unity_FC/1754 : FAILURE in 1h 34m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/3/check/DellEMC_Unity_iSCSI/c0fcda3/DellEMC_Unity_iSCSI/1784 : SUCCESS in 1h 34m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"621752c5a2c969782df0ea46d869d63e497013f5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-25 06:19:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/293/.","accounts_in_message":[],"_revision_number":3},{"id":"ae411ebb36516c086ab03d1eac8465452b2f84aa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-25 07:53:23.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-916467-3/FC/ : FAILURE in 29h 31m 41s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-916467-3/iSCSI/ : FAILURE in 30h 34m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"64f2835534f760ebc5c2cd8a8c98e49f14407d96","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-25 08:06:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/67/916467/3/ : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":3},{"id":"4a70c64dc77a15f493c663ec2bccd0f8e5fe8249","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-25 13:02:12.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"988caa34722008e869768be7cd0b7a3f33bde548","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:23:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee84c356c9cf4b26b5416ece39683b3a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d1340ce7947421bb11a17d1eef98285 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f7491179db6478fa05ebf374b939c0b : SUCCESS in 17m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99a3058b58a6085d83a3de0956030da31315fdbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:16:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23dbb1d9b0a14cc798aa2075adadb488\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89567d27bff04e3291060883f99e3785 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f71429347284607a8f08443605f2cdb : SUCCESS in 12m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37d1d6d7d52143cd80bb1a9f1e9faba1 : SUCCESS in 10m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c6424e4105841308f6a05c4f2cbf8ed : FAILURE in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2afeb65498034c8a86d0e4101cbe5f42 : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7865190c1bca4f4d903d869b3e0084ae : SUCCESS in 1h 11m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f32d337d4e024c1fb1a6279564904c6a : FAILURE in 1h 26m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a26bcddb22844e98034d81d48c70450 : SUCCESS in 1h 01m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/68d73b5c06f84507bc86d934bea6acae : SUCCESS in 15m 06s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/024ec41c327b4580b584bb9dce28779b : SUCCESS in 13m 02s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5fdf5ba6455d40588708bc8fd9179929 : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9bc9640914364db8957809e00b9f72e6 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9913e357c9e34d3ab32309694c151fd3 : SUCCESS in 6m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/14bf9dd6a20a47ddb032933c2fdb4380 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c544152207b64a02b8c3a8511172e079 : SUCCESS in 47m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d10cc0c8ba9c4cfeb53362dce94601f1 : SUCCESS in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c220c35af123459faaad12c4de6fbd8d : SUCCESS in 1h 58m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dda541957e3743cc80ea8a092696c68c : FAILURE in 2h 08m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10974b360fab4187b7e1bf5e9243f2ad : SUCCESS in 1h 32m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/31a84018656c4f8a8f98d59184428c30 : SUCCESS in 1h 32m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/50145f43335e49509a775012a5b09034 : SUCCESS in 37m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7d29b8d9b1ae4f218a19bc3450d64cd6 : SUCCESS in 1h 11m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/41ee4d2ce0584dc8880eb8877317642d : SUCCESS in 1h 41m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3af96329c76845668fed1b1602f70bbc : SUCCESS in 47m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc36cf438d454922b12fc57edd50e21a : SUCCESS in 1h 31m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4314a3c1f94b40b6a93aeb8e0b38a931 : SUCCESS in 1h 27m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/737a4993a5b84c03b21b0b0067d9ef7c : SUCCESS in 1h 42m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18135bc777d44397a63bae9007c9a114 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":4},{"id":"ddb6acee8da6f03c428768ccd1d5b1f900358fa1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 17:04:10.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916467 : FAILURE in 36m 06s","accounts_in_message":[],"_revision_number":4},{"id":"c472a34a50b76cfce62aea7d75a084bc76f0de6a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-25 17:25:25.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/66d93b207b3e47a893392eda4c7c1e8e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7c1af0505fba441c85e116f5646b786e : FAILURE in 1h 28m 07s","accounts_in_message":[],"_revision_number":4},{"id":"9f1d72e36627afd09dafb09105a2de310dfdc214","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 17:33:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916467 : FAILURE in 24m 32s","accounts_in_message":[],"_revision_number":4},{"id":"72406838555333cf3badc1f9471272f23583c986","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-25 22:14:01.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/67/916467/4/check/fujitsu-eternus-dx-iscsi/78da4ec : SUCCESS in 3h 04m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/67/916467/4/check/fujitsu-eternus-dx-fc/55ba0ed : FAILURE in 3h 16m 20s","accounts_in_message":[],"_revision_number":4},{"id":"a144e56e68913e84f6c0dfd716cdc660115c541e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-26 06:59:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_Unity_FC/8521157/DellEMC_Unity_FC/1775 : FAILURE in 1h 26m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_Unity_iSCSI/69d298e/DellEMC_Unity_iSCSI/1804 : FAILURE in 1h 55m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"f35083df191d3d7e4385a76f97eefbc687aa41d9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-26 09:06:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/312/.","accounts_in_message":[],"_revision_number":4},{"id":"02433fe64bd93d9b397e297bbc729ccd46231ceb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-26 16:03:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"02fc61b5f32fc0db9ae4f32919ca0c726d69fd72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:39:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/594f4e2829c44bf29fe64df32f4d4902\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20e5c8069e674860b43c13abf767fed6 : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d2919781e26743a1b88275bf4510a0a4 : SUCCESS in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f27c56c106ddf17a736a41a65a46e78881e4d9bd","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-26 19:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/916467/4/fc : SUCCESS in 1h 01m 56s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/916467/4/iscsi : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":4},{"id":"c10e76235a29b3092c1382e71e0929087cc65d48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 19:06:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c6159d5a3bf45f7af72355bef9dd773\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bbf56d5b7814337965dd8d29e426e85 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9f737afa70044a8b77be973760b94a5 : SUCCESS in 11m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93ba6efa7d76410d84d3abd6b3107548 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0c0ceaef6f649fa96078b81b88b4e83 : FAILURE in 12m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f3d9e6e809412b845f3b09ebaea696 : SUCCESS in 13m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c36661cdd7fe44c7b1271017eb724ddb : SUCCESS in 1h 03m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fd565f6b4cd41b8a5d01fcd57fcd6c8 : SUCCESS in 1h 23m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6c12ee4ad9d462aafc44be395fbf22f : SUCCESS in 1h 01m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d11826027b32442885891a55b5eb12ff : SUCCESS in 15m 07s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5e1c84a86a954385aa523f6c933125b8 : SUCCESS in 14m 38s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b3c4d84de1de4e4e8538883b145c829b : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb6ce102b4344a38ab242a34f23eff52 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1123d7c6e4a4076b8a433aec219622e : SUCCESS in 6m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/19ad8a5da7e24dbc955beb75e3970939 : SUCCESS in 8m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/389eaffcb1434737a26ae95bb4aa34b8 : SUCCESS in 35m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f88b9b16e01425999664a069a73e279 : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0dd78eca38644da09f6767b692e4cf60 : SUCCESS in 2h 23m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d784629e370e4485a802f20c770f290c : FAILURE in 2h 48m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d4691cce90c14ad4a33d78a220cac523 : SUCCESS in 1h 57m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/05964f57bf034cfa95916f0f512de4b8 : SUCCESS in 1h 47m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b8c6aaa8345c4756869c139b2eedfb22 : SUCCESS in 39m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da24b64ccd17481f984d7254c32df715 : SUCCESS in 1h 13m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8882df744d554c329eb614d687c546da : SUCCESS in 1h 33m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9195a8bca89a458aa5fe4cb83942e76d : SUCCESS in 36m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7211c44419344c93b434d10e0bb0081a : SUCCESS in 1h 24m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1bab3f6d882c4274971c72d567c75a59 : SUCCESS in 1h 20m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90eb89058c94438b841554c9398b708e : SUCCESS in 1h 37m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05f4c246c31d418e898595cc79291bfa : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":4},{"id":"e61ef13a23d10f67fa190d49d0d1078a58036304","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-26 19:13:11.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_04_2024_19_07_05\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"02d832ab119d6aefa2c94c6308f60c2f2d3e4618","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-26 19:28:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/916467/4/upstream-check/cinder-tempest-plugin-ontap-fcp/e768749/ : RETRY_LIMIT in 3m 56s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/916467/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/69d0cea/ : RETRY_LIMIT in 4m 12s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/916467/4/upstream-check/cinder-tempest-plugin-ontap-nfs/5b3f6ec/ : RETRY_LIMIT in 4m 04s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/916467/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/b3448ef/ : RETRY_LIMIT in 4m 09s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"b5f73a9a302e058d31b630fdb038b42e10dbe23f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-27 00:09:55.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_PowerStore_iSCSI/23ecee4/DellEMC_PowerStore_iSCSI/1283 : SUCCESS in 1h 11m 31s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_PowerStore_FC/97a0734/DellEMC_PowerStore_FC/1454 : SUCCESS in 59m 21s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_PowerStore_NFS/5b7d64a/DellEMC_PowerStore_NFS/973 : SUCCESS in 1h 13m 51s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_PowerStore_NVMeTCP/75dee42/DellEMC_PowerStore_NVMeTCP/834 : SUCCESS in 2h 01m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"ff2514935103a938204fb776b82486d0223495c2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-27 00:26:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_Unity_FC/3acc190/DellEMC_Unity_FC/1785 : FAILURE in 1h 21m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_Unity_iSCSI/a131b26/DellEMC_Unity_iSCSI/1813 : FAILURE in 1h 49m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6fe0dee54bfb62c74b5a95185d38295a89fc3d0b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-27 03:40:51.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fbc26da6d0964f718aaaed2e7b7b25d5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e5f0b150a3a14df3ab4b36b9c7f8d363 : FAILURE in 1h 26m 32s","accounts_in_message":[],"_revision_number":4},{"id":"3bd69c6b7e41d1f0018cb35a2c16283fa63c343f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-29 10:45:41.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"d93261f10fcd79a49737534e8e02c8981c2c90e0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-05 12:13:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/67/916467/4/20240505161748/NEC-ISCSI : SUCCESS in 4h 18m 21s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/67/916467/4/20240505161748/NEC-FC : SUCCESS in 4h 55m 36s","accounts_in_message":[],"_revision_number":4},{"id":"fe210d61dd5ba958e5428d496f2546d7f4960c7d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-11 02:30:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_PowerFlex_3.6/3a75568/DellEMC_PowerFlex_3.6/1275 : SUCCESS in 1h 41m 59s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_PowerFlex_4.5/23f1173/DellEMC_PowerFlex_4.5/209 : SUCCESS in 1h 49m 38s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/67/916467/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/63f475a/DellEMC_PowerFlex_4.x_NVMeTCP/150 : FAILURE in 2h 54m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"0b52b64ed5502f208bd2c8436c701317f78711f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5634091861a2f47e9653d5bc5460586ca6c60e6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:19:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de0d960630b547558f89a767e735e16c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/835dee7aa2a946148cb440eed6a981b4 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5262444d511b4cbaa4ee5d86bb711722 : SUCCESS in 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"05f7e1549490e6231f2d23746fa27beb16ac9a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:30:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4648b142c97a4b9eb322791db968ed00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825bf2c2a3a5400e85fb0cdf23771f21 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c185989c1b834fa286c4fde4e395a3a7 : SUCCESS in 10m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d831da090e22423b93e1634004930cfd : SUCCESS in 10m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88373898a29f45a98bb9227602425c87 : FAILURE in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a2bd3c0b7c4310995f9e4e916eb7dd : SUCCESS in 15m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/d59bdedf01f948fcba1c7736f129bcca : SUCCESS in 47m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7358f3b0a8394827988a31c220982564 : SUCCESS in 1h 59m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca73efc121214e2f96016e281b853363 : SUCCESS in 57m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4ae04b88285e4ffdbd4849303d8ad19c : SUCCESS in 15m 37s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e4e5804648e743a897264f0e455d58b3 : SUCCESS in 12m 40s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/28ff020080984185804daf271dcc696b : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/27bad688fded46c29b507770d27f156a : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/095e6281e50c49dc8f75162c88ac7cb7 : SUCCESS in 7m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1014f9650d2d4f69978ab813923f801b : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bba819dcf0b649a6af16fbf67fa55dd0 : SUCCESS in 44m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b7c5487757e4bdebac497c6b03aac03 : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5e87cdd446314383aade157b424a5999 : SUCCESS in 2h 00m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e1232f6c2584b438e2c19852474a909 : FAILURE in 2h 01m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6eee7c5b75f24ddbaa7ec6744f8a9b7e : SUCCESS in 1h 23m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9769c76b97254cbbbc19ca074a950432 : SUCCESS in 1h 58m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b206d6fb3bf943cb92abddadf6832de1 : SUCCESS in 33m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b51d5741ce6242c39b099b7a0c9c1632 : SUCCESS in 1h 18m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/239e545791334153ba949e4cafa83699 : SUCCESS in 1h 34m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b8eb229e343d4f2d830bbc807f9cc788 : SUCCESS in 44m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6bdfaf6e0df84bbfa38027549d872b7d : SUCCESS in 1h 16m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4d6e3ff5c8e54f438023c8dcd86e9153 : SUCCESS in 1h 12m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4daccc40be634e3bb35bd3f8a3d20419 : FAILURE in 1h 35m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b9bfc0851a8412c90d835d3eaf1d31f : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":5},{"id":"7654982828f83be340d128f2c4c712c2c438a224","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-14 02:49:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916467 : FAILURE in 28m 47s","accounts_in_message":[],"_revision_number":5},{"id":"9b1340d6fe6480f1bded198a58d6a6687ed1a2fd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-14 04:34:18.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8c1fbce2914449ef96d13d8a9cc94e85\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8c5b24de937f42f3a7fd0f8c574976c5 : FAILURE in 1h 30m 14s","accounts_in_message":[],"_revision_number":5},{"id":"70059141dc625d9fef84c3cab5381b56e174e2f5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-14 15:50:48.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/378/.","accounts_in_message":[],"_revision_number":5},{"id":"460b719b423251292af835b2bd870232ae093e18","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-15 00:43:51.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/916467/5/check/DellEMC_Unity_FC/0de9108/DellEMC_Unity_FC/1857 : FAILURE in 2h 08m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/5/check/DellEMC_Unity_iSCSI/999d37d/DellEMC_Unity_iSCSI/1885 : FAILURE in 1h 56m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"63419f83c5af5ec338fb0b7f2905a20f1fe70358","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 08:20:45.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/67/916467/5/check/DellEMC_PowerFlex_3.6/14426b5/DellEMC_PowerFlex_3.6/1339 : SUCCESS in 1h 38m 42s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/67/916467/5/check/DellEMC_PowerFlex_4.5/6e3c768/DellEMC_PowerFlex_4.5/264 : FAILURE in 56s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/67/916467/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/0ae61ab/DellEMC_PowerFlex_4.x_NVMeTCP/211 : FAILURE in 1m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"d4155fb5a864d82a42bda245695353eccbaf5395","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-15 11:00:08.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/916467/5/fc : SUCCESS in 0h 42m 10s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/916467/5/iscsi : SUCCESS in 0h 54m 19s","accounts_in_message":[],"_revision_number":5},{"id":"2f056a668f681a060b5e583840fbd9619ece6a2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4772ab60fef94429964bfbb070c2620352ac5043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"d70477e76cde280271bf104993e1aff99808dcbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:32:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a36d2a2379f43c6a0dc7bd9ce1685ed\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08166185ca8b435286b8cd4071c6d76e : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/196ec060aa6b4b53aa1e1556289bb507 : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a98ffcf57de406b97c76b60afbfa36947fcbac1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:38:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a0e6df13789485dbefec28018370f96\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aef55e863de43b2b06ef12676c3f107 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2da93632ab38407aa993a27bf62a5b7a : SUCCESS in 11m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a74f0162b8a413cbfbdede3a86bb785 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/400744ea45b142b395acc31f7d741db9 : FAILURE in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afc189b49a8043a08305138c7e9a9749 : SUCCESS in 14m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/24f11cfa3c494bdf91cfdd4eebca5a86 : SUCCESS in 1h 00m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91282aba932a4ce28220ad623e212302 : SUCCESS in 1h 39m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80968760d2664bfcb4fcd3d3b047acb2 : POST_FAILURE in 54m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/60392fc9d0054b73b80ee06210e4b034 : SUCCESS in 14m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a3fc867cf688485ab8b92e223f51487d : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f8b82623175f48eebb70cefc05507a35 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e866f42f53f44faaa69043495f741d89 : SUCCESS in 6m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fef6b468dbf244ff93bb4472dd8aa5c2 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9a878c34221844c0a433fca51391ce47 : FAILURE in 49m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cffd73183ba642be87b57b950dbdd0e1 : FAILURE in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f1b1ea9092e49d0bdc84c1ae8371043 : SUCCESS in 2h 09m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/369f2e101cef4a0f830c21bd39caffc7 : FAILURE in 2h 20m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e06b4d8690b54627b606c0bcebdf148f : SUCCESS in 1h 48m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/18746a84e2cf4856a5c4b58bda911dd6 : SUCCESS in 2h 10m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4371852f3ea94c64a312711933180423 : SUCCESS in 40m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/32f11b515f154df38addb26d75933b2a : SUCCESS in 1h 18m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/32e9cfa6d40f40bd867b4b35c2b4f334 : FAILURE in 1h 46m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/36b6a1573fd24ef999b48e044e1d0c36 : SUCCESS in 49m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a804c5c57804315964aa3f8b0899eb6 : SUCCESS in 1h 27m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/378b11b547384c61bba5ad43619d71cd : POST_FAILURE in 2h 17m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2c5acebb7fa44dc9013735d61f2d962 : SUCCESS in 1h 44m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85020ce37b9c43a993e228d4873186b1 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":7},{"id":"16e0fa37fe558f134ca0c8c173bc957f13203919","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 16:06:07.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/84c3536d58fb484093436a2f88926301\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bc77ce21a3cf4f51ac023d5454659cc7 : POST_FAILURE in 23m 06s","accounts_in_message":[],"_revision_number":7},{"id":"286de8e44b9436e24bdc01e17a997e79e9d8d93d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 16:06:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6adc360ac9c1f422211ded109572e644df4a0853","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:25:04.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/874419abfef14438aaeac5db35da0cee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96323a8cc4714c7b9198a16ff085ab23 : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2780f0e0c854a2a87034cd99160a95b : SUCCESS in 16m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cf7eed007e6620d0ef7b07e57544e10b71021049","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 17:45:46.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ca30bb2bb352478a8b4018d973df6e73\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/71d912ccfe9e40bfa8c1967d6e8eba1e : POST_FAILURE in 22m 33s","accounts_in_message":[],"_revision_number":7},{"id":"3761a88818270b5397e4c62c006e809e391fa0fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 17:48:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916467 : FAILURE in 23m 20s","accounts_in_message":[],"_revision_number":7},{"id":"812fa0feef7912e8a8eb5c118aa29a2f4d068f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:21:51.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c534d0a9c7e4b41b34e28b0fb73308d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f218839c73e3408eb89af539fddd63f3 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07e7a4cdb5af4f09b427e80170c129d1 : SUCCESS in 11m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0df0871cac74479a9434377ea1c4eec : SUCCESS in 12m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/763981ac68244500b1930d5223dc49ff : FAILURE in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00575553e7b14b89aaab514720ac0cd2 : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/8d09dd7552384bb6a7ef3cc6ee38847a : FAILURE in 45m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5bb05e05e66548c4aa3a3d80d54b5c8d : SUCCESS in 1h 32m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46ff1f2571d34f829359dcc2ca75684e : SUCCESS in 54m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6b6b2343f03a4adc98510979e8dd1c58 : SUCCESS in 15m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5d49249401284871bfb085eb96626dfb : SUCCESS in 6m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/558846d51808420282a93b4493ec6ad0 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2588465fd8a42a3993b1cd5e469c197 : SUCCESS in 6m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/010c74ea852f4707b729d096ee3a0e49 : SUCCESS in 7m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0f299a598a5a4924acb506a569e38997 : FAILURE in 46m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/002fa418562245838dd86b5e643e76f9 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d22823393e5d4d4798a64c5355fd558e : SUCCESS in 1h 59m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3ebda7b00e304f0ab50fcc8dd5f09179 : FAILURE in 2h 04m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/788c804f3fb44512865deb5b04fb033c : SUCCESS in 1h 32m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c22d7cd2c70f45f495eb90a74412741e : SUCCESS in 1h 53m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3ba05aaadd874e81ba4d24a1d1fe4836 : SUCCESS in 35m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/470a108cb9e04574aa9184f06b08976c : SUCCESS in 1h 00m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0710675045446609e013cbc168914ac : SUCCESS in 1h 40m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/32408bd3662c43fbbe743a713fbb8a50 : SUCCESS in 42m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1cac6f9bafae47b39393b8e5000951f2 : SUCCESS in 1h 16m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/91173b7a23bb4911bf27baa3e0ea16e2 : SUCCESS in 1h 21m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d7c868086374684b11dccabd4eea8dc : SUCCESS in 1h 43m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5e303710e28413ea475422aafd76d36 : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":7},{"id":"e12d0b5984db9b853e5c7bd37dff3654874926e7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 22:42:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916467 : FAILURE in 36m 13s","accounts_in_message":[],"_revision_number":7},{"id":"37f2f0ba1ae630f693e2ac0370e2e296299dc1dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-21 12:59:05.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/440/.","accounts_in_message":[],"_revision_number":7},{"id":"720aa75cb8b2a8c4fb3717cdd8f30aa889dd7b99","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-22 01:19:24.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/916467/7/check/DellEMC_Unity_FC/1d1a09b/DellEMC_Unity_FC/1917 : FAILURE in 1h 49m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/7/check/DellEMC_Unity_iSCSI/ff14444/DellEMC_Unity_iSCSI/1945 : FAILURE in 2h 03m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"905244f3fd95e92b877bb9b1aa5365090dcd4f60","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-25 03:04:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/67/916467/7/20240525040335/NEC-ISCSI : SUCCESS in 5h 03m 22s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/67/916467/7/20240525040335/NEC-FC : SUCCESS in 8h 01m 17s","accounts_in_message":[],"_revision_number":7},{"id":"432d36bdc516929477abfe1f81bde5d102411e3f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-25 22:25:43.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/67/916467/7/20240525230325/NEC-V-FC : SUCCESS in 3h 43m 30s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/67/916467/7/20240525230325/NEC-V-ISCSI : FAILURE in 8h 22m 14s","accounts_in_message":[],"_revision_number":7},{"id":"cbb2b80fb131bfd04f2cec00f0288271581c0e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6dae9d43c619aef58a78fe429e37d4cb9764772b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:45.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/593/","accounts_in_message":[],"_revision_number":8},{"id":"acb0d373df51740e500a129a9087c64da89820c5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:47.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916467/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916467/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"7a1938751a85919f7e7d78fb5ed6f8daa8765f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 17:50:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1ebcfe2ae0f425b8f870b0a014e7784\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/967e8e1cbe584a9583b3852818955510 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a9aa74eb21147e08d0cbfa586e8edc8 : SUCCESS in 12m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0a2e74b8d6440e394e0332c9085085b : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d30a75ab373a457687c3fc8ae08a4884 : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0680451b840d42b4b50930434bdd4c70 : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/055cb03a1c004bc5bfafd841eb2d46b5 : SUCCESS in 1h 14m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bfd765eff7a04a2ab16c1504b25de255 : SUCCESS in 1h 29m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8cba8912aab4f308e09f5d85e4a9c0d : SUCCESS in 29m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fd5c9109c7a44488a292369518f72853 : SUCCESS in 16m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ce3f932406db4e5fb0634436e35a9cb2 : SUCCESS in 3m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a5012db32c0349938237ca5f8e2f1581 : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/959c2ab229ac4f618bfd3b6506f7f210 : SUCCESS in 9m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f96dc9bd8cb34c12b8e5e41a29e81c97 : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/967e705aa3f447b490ab90529d77acd3 : FAILURE in 25m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/693513ef56984898b0d4fb29ad151a41 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2ead95b3e11e45b18b964ddfba4cc514 : SUCCESS in 1h 54m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a09725170f274b54bbf5b311d92d35d8 : FAILURE in 2h 28m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/76c831ccf2dc42ac875b3ae01cb8b45e : SUCCESS in 2h 20m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/78e6ffcdfabb4f50bc157549e8c008d6 : SUCCESS in 2h 12m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d7a2cc94121d4b8a96b3b08a37b72a1a : SUCCESS in 31m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0f63d7cfa8b44e638aed4b4c444bc748 : SUCCESS in 1h 12m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b97f1ee4167c4e18b092d355bc8c1a0f : SUCCESS in 1h 38m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c0024bccf7e749fd9a379436813ddaf6 : SUCCESS in 40m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/89160d9607884e66a175f7ac58b0b868 : SUCCESS in 1h 40m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7ae35c6dfe0c437ab9698302a0c41080 : SUCCESS in 1h 41m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/939143d5427d45e1b39b18828906bf8e : SUCCESS in 1h 57m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d37ec67b993a4b849d8d6898e78b834c : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":8},{"id":"cc5bb213432c1db4a733e2198a14f442584ed1d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-18 17:56:06.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2024_17_50_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"e6206b5a3755c0e9a044c3db004127e5f3b4deea","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 18:59:40.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/916467/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/78b03a0/ : RETRY_LIMIT in 21m 22s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/916467/8/upstream-check/cinder-tempest-plugin-ontap-nfs/11a5ed6/ : RETRY_LIMIT in 26m 09s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"d99581b7cffafeec39d2e15bb07c6cffe264b1a6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 20:34:04.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/40c34ba9b74f48eb860caf02e1d8141f\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/497a903fbe294bb1b99fea07dd8c6c6e : SUCCESS in 47m 06s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0a443f2b5bdb47d495dd516a86b6e532 : FAILURE in 44m 01s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/60067d2260a64538926b1a1d0b946921 : FAILURE in 2h 10m 23s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6b7e8bb5cda74a55b929ba4535754522 : FAILURE in 42m 20s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"10237cb629e4ed4f996550756fbeac01812256d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 21:13:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1181a07580b54293ba5cce18d2a06930\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6fa5a1eb47924bc9b9d041e807d61990 : FAILURE in 19m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bcff3f72f1164216b4eab1d01f1f3f41 : FAILURE in 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f191ab0eaa9cde9a03556f77480c4fa63bf1daec","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 23:17:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/916467/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/105d849/ : RETRY_LIMIT in 19m 18s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"21824c449b0daa93c95862654b385d86e6186aca","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-19 01:12:40.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916467 : FAILURE in 20m 28s","accounts_in_message":[],"_revision_number":8},{"id":"0868bad02fb3d53cae992dee5c84e686042afa1b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-19 06:25:34.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_PowerStore_iSCSI/16783ee/DellEMC_PowerStore_iSCSI/1752 : SUCCESS in 1h 26m 52s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_PowerStore_FC/bc2ea11/DellEMC_PowerStore_FC/1950 : SUCCESS in 1h 02m 14s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_PowerStore_NFS/bff6f3d/DellEMC_PowerStore_NFS/1445 : SUCCESS in 1h 13m 35s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_PowerStore_NVMeTCP/78c6784/DellEMC_PowerStore_NVMeTCP/1305 : SUCCESS in 1h 54m 23s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_PowerStore_NVMeFC/fcec9de/DellEMC_PowerStore_NVMeFC/269 : SUCCESS in 1h 15m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"447a57a9823401c8cadd13b7a42ed3528545a47c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-19 06:50:59.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916467 : FAILURE in 20m 04s","accounts_in_message":[],"_revision_number":8},{"id":"7ebb89da8197c470efad84ae17f55618e6a7ad13","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-19 07:19:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916467 : FAILURE in 21m 20s","accounts_in_message":[],"_revision_number":8},{"id":"a3e3cd8f157c9da24e84e7b46ec74a6faefba138","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-19 17:24:22.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/44150b4fb4ee4a75ac3dd6deeaed1a0e : FAILURE in 1h 15m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/79526df9f0a346128b03fea374fce703 : SUCCESS in 1h 29m 18s","accounts_in_message":[],"_revision_number":8},{"id":"8302a9f2d097ea95fb7444869c511a035c38ccd5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 04:01:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/569/.","accounts_in_message":[],"_revision_number":8},{"id":"ea5252be8cfec4cf9716157fe04e6c0aa7e5b462","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 16:19:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1157/.","accounts_in_message":[],"_revision_number":8},{"id":"8894ca2adeb470c68e8d3a95268ac0b48c1b7343","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-21 23:51:01.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_Unity_FC/1215405/DellEMC_Unity_FC/2637 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_Unity_iSCSI/7fc1d8d/DellEMC_Unity_iSCSI/2646 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"85c07b16783c611ffe670f84cfac79bd1febcabc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-23 17:08:48.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_PowerFlex_3.6/79d3a6a/DellEMC_PowerFlex_3.6/2131 : FAILURE in 2h 32m 31s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_PowerFlex_4.x/a788f95/DellEMC_PowerFlex_4.x/1015 : FAILURE in 2h 13m 00s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/67/916467/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/8dcadc7/DellEMC_PowerFlex_4.x_NVMeTCP/955 : FAILURE in 5h 54m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"171a5bc8adf23532b0cd911b073bf670e17fe2ae","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-24 07:39:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/67/916467/8/20240924125014/NEC-V-ISCSI : SUCCESS in 3h 48m 23s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/67/916467/8/20240924125014/NEC-V-FC : SUCCESS in 3h 49m 32s","accounts_in_message":[],"_revision_number":8},{"id":"fc309a0fe418463f5f9d9c2097e2a823dc282b46","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-11 18:33:22.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"7fc84c1b5bedf42de149538cb3af4c5213077b75","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-21 09:25:53.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/67/916467/8/20250221165740/NEC-ISCSI : FAILURE in 52m 18s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/67/916467/8/20250221165740/NEC-FC : FAILURE in 1h 28m 11s","accounts_in_message":[],"_revision_number":8},{"id":"13790a35e0bd55fb824c681d61f0fcbd16b10b59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"6f8476493765dcc1419fdeeebffb8a9671920057","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"5ec99009ca1168ceb566d159a9cea73fadf89bba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 12:54:03.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"61d5e702f70ed68faf674c74cf53c614b953138f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:54:48.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"51a7a3195f63fbba21fa6c5017ba06b5276873b1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 12:54:50.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"643399dcb3eaf7c6f06f86127ceca43370c691a9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:20:01.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"5cfb9b321194f92881dc2c580a6c58d6ddd51db5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 14:55:17.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I174c3bae39625b74e4b416e3559eb79ad8d34948 .","accounts_in_message":[],"_revision_number":9},{"id":"965a37417d60105b14c2e9add444911de5c249c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 15:24:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848b79f7916c4e71bf5a5ec8c1805466\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91ed970aa95a409cb9d028c67af7bbc3 : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa38ee8b687848a18d446caa0ea4299f : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b527e02f39004f78854446786254f0f3 : SUCCESS in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d8093dc4afa93b092e05a895d22a99ea98eb9fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 16:34:04.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9e82bd944554a098d8d95e32bc42145\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b87f6d5a4a4fb189504aa0684752d2 : SUCCESS in 8m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39632dac1fb04e92975d644e77a31805 : SUCCESS in 10m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07b9cd2f62d349269c7a1743a891e683 : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1bac11c0594e57845984a9f09559bb : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a9ce0ecebee2473f9569e6f5d3c17f7f : SUCCESS in 59m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/754f800a5b9c410d89f0b04aa90da136 : SUCCESS in 1h 02m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd19616e77b5486dae5e9913ba792433 : SUCCESS in 1h 49m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d50093fc6774ec689faaa34a936d141 : SUCCESS in 48m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a2313ee26e9441338c7711ef74630269 : SUCCESS in 22m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d3619ed5f3b14c8195e65de4d02d0a5d : SUCCESS in 3m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2df27bcaa8b64e97803db67e4b779326 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b3d6831204f645e1b4f7e307218f8b7d : SUCCESS in 5m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/00df7b879fd54073a5e3061f7e3b94a5 : SUCCESS in 11m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cbbb22291341497bb373c1b756e26023 : FAILURE in 27m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc387d8bb2fa4fd7bc1afc87c47e0a59 : SUCCESS in 4m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/294b2f33e21e4fcf876e32dd178651e9 : SUCCESS in 1h 08m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ec41b1a660c94b118d5484d7b975939a : FAILURE in 2h 49m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8768e0fe500040ef9bc19e5eb9439003 : SUCCESS in 1h 48m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/65d222ed08fc40e6ac68b0e1429f10ba : SUCCESS in 1h 52m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/247af186d1494ba98f78d7bf6a726ef2 : SUCCESS in 29m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae4b6d977aa4458789f67da698177298 : SUCCESS in 37m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/56769ac617f54a76b2d3e7c824c29716 : SUCCESS in 1h 29m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4568aa8fd88c41c1b797f7478dac6810 : SUCCESS in 37m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e0e60b7705346ceb344ac220aeb97ce : TIMED_OUT in 3h 34m 40s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a4f45ad8966f4ca8a47cee0d2ba5f9fe : SUCCESS in 1h 15m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6c5b3d966284e068a874ac2b9499cc9 : SUCCESS in 1h 20m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/066a3cede527439d98c9032270152ed0 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":10},{"id":"c699e93ac50136642a4260464d322707d60c48d9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 16:59:23.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I174c3bae39625b74e4b416e3559eb79ad8d34948 .","accounts_in_message":[],"_revision_number":10},{"id":"747426f218d47d44e85659212854e135925eb39e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 04:03:54.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2260/.","accounts_in_message":[],"_revision_number":9},{"id":"e3d433573a65d545e4bfd5612244eeeb7cc95b11","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-25 04:10:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b5e4b51172604efdb403ab3fb71cbd36 : SUCCESS in 44m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/02a4c50b1f0644a992a612dc0fe29045 : FAILURE in 52m 36s","accounts_in_message":[],"_revision_number":10},{"id":"8d1cd55260d0ab6b7118494feda998fa12cb3be9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 04:28:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2146/.","accounts_in_message":[],"_revision_number":9},{"id":"d0b38eda94ce5fe80dfa43c7cc286c3f3df7c213","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-25 05:06:45.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916467 : FAILURE in 36m 26s","accounts_in_message":[],"_revision_number":10},{"id":"bb4ae4f589db8002584e94c608d014c695f519cd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 19:28:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2278/.","accounts_in_message":[],"_revision_number":10},{"id":"50d7bf5ae3cc1a5963fa7be0c9dfc26307f0204e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 20:13:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2164/.","accounts_in_message":[],"_revision_number":10},{"id":"0c92a96f5f9da0114863a298e514abbbf7bcc8c3","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 15:08:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"be57c669667674448a38d3b8c516f4f53d946ced","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 15:53:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19bdf2eed2cc4cf99b2fccb27a837b13\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c01f329282d4216ae07afd46fccd9df : SUCCESS in 23m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/732ff4215a3646a38c006bba5197f8c0 : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4acc49ba25994aaa83d8d09f34231f90 : SUCCESS in 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1a98c5630dc0dfdb8e5efa3cbc81f016099745e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 17:48:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d08a2ce3c99c4938a7b2fd18f40d2911\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77342ced14b445cd9bbe67e56b336353 : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/351af09037e444bd81fcf8992d1cb8f2 : SUCCESS in 9m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a425208aa904baf9c221209d5a6fdd2 : SUCCESS in 14m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d82d9443832445db9266142c459ff209 : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/f10516b1689c447fbae22ad0e2c4446a : SUCCESS in 59m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b37e7df059cf4aaa99e81a1a5de7f965 : SUCCESS in 1h 00m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed56b8a430f04f11b2f023989043dd48 : SUCCESS in 1h 25m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5180729676d495db0c04bb7e2497752 : SUCCESS in 49m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/93fabd41e158432b9da0de2a5eee529a : TIMED_OUT in 41m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a9f74c814381483896eb003f12295f23 : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d08ff591640b439aa61446bd30b12749 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b9a2c652a60f4a0eb5e6ada306bc0af4 : SUCCESS in 5m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9d00bacefcae42dd863a337880026529 : SUCCESS in 9m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/690ebfbe921c481390b00e56a76b606f : FAILURE in 35m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19295f12d74c4b2f875a3511c7d7a0e2 : SUCCESS in 3m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/255eef52aba94f8fa1eaa87aca71e513 : SUCCESS in 2h 13m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ccf2fb34baa44aca92ab214a3862f6ad : FAILURE in 2h 31m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f066b2b360f74db59de5720bb90a96ff : SUCCESS in 1h 59m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c7183fbc7d7e448b8da295cb3e6a78d3 : SUCCESS in 2h 28m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0f52cef5fdfa4a2aa0613198d36f7fbb : SUCCESS in 24m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6939f3864f9c44ae8c7a4b51a170d943 : SUCCESS in 1h 12m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7729824d84a8440e99f9e3111f01e07d : SUCCESS in 1h 45m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d6166e472db241c6945f93211361ba0f : SUCCESS in 43m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5744751f1f574fefbec835bea0b0bb13 : SUCCESS in 1h 55m 23s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0493d195e14b499fb8dfb0ce12ea4aa9 : SUCCESS in 1h 40m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/068a3466f3e24616b3bab5c3b9860f3f : SUCCESS in 1h 49m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e8546643b6749ccb5f3182ba124bbd7 : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":10},{"id":"87d7855e759ee40b83057c3ca2692123f9ac1eb7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-30 18:03:51.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_10_2025_17_57_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"61fcecf635e5b2e10ff427dbdbe854d850e00c86","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-30 19:02:27.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d0169c3b1b70418aa8395bcb3284a966 : SUCCESS in 48m 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7e7e2767cff84c34819cec2c12d3b4bb : FAILURE in 1h 05m 35s","accounts_in_message":[],"_revision_number":10},{"id":"5ad2eba4d493de342d688502fe3a2093f25083c3","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-11-11 17:25:41.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7bdf332faf11ca7d47af568d69ae278fbcfafe9f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-12 15:27:40.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"dca8c527269595c8e6952d5b8332adbf648aca9d","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-11-12 15:27:57.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916466 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"0fed23608904e74347c186f715fd896eaf97e373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 15:28:45.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ad6ecbc695cd3e7bcf733b76c6f710e8ae24bd7f","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-11-12 17:02:07.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916466 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"e845d03bc69ed8b32a7fdcd1cf38bd866c3ecd23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 17:44:12.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff06e5bc5414493aa4fb70e211b9fa12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a00896a9bac4b7485347fc76e223bb7 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c3df6ff8ae14ee9b80a06d726a70702 : SUCCESS in 11m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfe5e2c7f18a42888f9d7d77885f4565 : SUCCESS in 16m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d91f17e448c4b948a2e390b53aed9ce : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/bef7933fc680450b91e6e610ff0eac4d : SUCCESS in 1h 01m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b55be54dc680457c92fee9912ddbff0b : SUCCESS in 50m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73e62a6beefa4bd99cf89aba75dedd00 : SUCCESS in 1h 25m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/350548723c3a46a0bc2c83622bf15a8d : SUCCESS in 1h 05m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a89fb23479b1411e94abf499f8c14f70 : SUCCESS in 59m 13s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5354d6d3e42498a848ffecf9836ddec : SUCCESS in 1h 08m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aeb4d9ee81b442ef828a83701e5882d6 : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":10},{"id":"e5045501ef11983c20de407903bfade5d980afc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 17:44:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"e2ecabdb84acb0f48abdbee0ee80c8e2d59e7742","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 17:45:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c3432603b004dd2a84e6f978118a94c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c423bb2be84925b60f5a0ac9ad4db5 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"66008dd93ce0105089edd537b4b1d82f66b3d688","revisions":{"919323e30d39e9c204ba9a98fcf1324381df9564":{"kind":"REWORK","_number":1,"created":"2024-04-19 14:51:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/1"}}},"commit":{"parents":[{"commit":"d3dd7a45f2943e390b4cb28a7b81a56fbc1f18b3","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3dd7a45f2943e390b4cb28a7b81a56fbc1f18b3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 14:49:25.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/919323e30d39e9c204ba9a98fcf1324381df9564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/919323e30d39e9c204ba9a98fcf1324381df9564"}]},"branch":"refs/heads/master"},"68bba0910d2925bfcf59e93bf5301c72638265a6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-19 17:27:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/2"}}},"commit":{"parents":[{"commit":"ce39c9925587691185be4ee63cb91120b9b1abdd","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce39c9925587691185be4ee63cb91120b9b1abdd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 17:27:38.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68bba0910d2925bfcf59e93bf5301c72638265a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68bba0910d2925bfcf59e93bf5301c72638265a6"}]},"branch":"refs/heads/master"},"f470a03134260472c135909052e094ba64118c7c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/3"}}},"commit":{"parents":[{"commit":"eda97b458b80f7ac5f75711fcfaf86b79a6825c7","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eda97b458b80f7ac5f75711fcfaf86b79a6825c7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:51:11.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f470a03134260472c135909052e094ba64118c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f470a03134260472c135909052e094ba64118c7c"}]},"branch":"refs/heads/master"},"cafa49a67449ca797091e6c362c0d696f02157b4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-25 13:02:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/4"}}},"commit":{"parents":[{"commit":"72529a04e00248d1c4cb1a765402364ae873fd81","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72529a04e00248d1c4cb1a765402364ae873fd81"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-25 12:43:38.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cafa49a67449ca797091e6c362c0d696f02157b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cafa49a67449ca797091e6c362c0d696f02157b4"}]},"branch":"refs/heads/master"},"952273d208d6bf2fb8dc6f2b49dd9c7e968de268":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/5"}}},"commit":{"parents":[{"commit":"ac1d629e92df54c5787d1f332082fc82f314e684","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac1d629e92df54c5787d1f332082fc82f314e684"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/952273d208d6bf2fb8dc6f2b49dd9c7e968de268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/952273d208d6bf2fb8dc6f2b49dd9c7e968de268"}]},"branch":"refs/heads/master"},"5d5bf44f526e3ea0a67981329833638d5969d6d5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/6"}}},"commit":{"parents":[{"commit":"47d833289ba7509bc10a5cb671f72a6c61a2567a","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47d833289ba7509bc10a5cb671f72a6c61a2567a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:00:19.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d5bf44f526e3ea0a67981329833638d5969d6d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d5bf44f526e3ea0a67981329833638d5969d6d5"}]},"branch":"refs/heads/master"},"92e561c775433d598f9bb0e303ed318f0af41f5e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/7"}}},"commit":{"parents":[{"commit":"c00de985a6bf3e87df3e0f83ade39c823a6fa954","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c00de985a6bf3e87df3e0f83ade39c823a6fa954"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92e561c775433d598f9bb0e303ed318f0af41f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92e561c775433d598f9bb0e303ed318f0af41f5e"}]},"branch":"refs/heads/master"},"5cafe2899f923ca36e3fffa0b25bd6deca2c5702":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/8"}}},"commit":{"parents":[{"commit":"56142eb218a06e8a2ec5f7f22aede29dc9211dfb","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56142eb218a06e8a2ec5f7f22aede29dc9211dfb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 13:02:23.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5cafe2899f923ca36e3fffa0b25bd6deca2c5702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5cafe2899f923ca36e3fffa0b25bd6deca2c5702"}]},"branch":"refs/heads/master"},"a995a2b21e638235574c002d5fa90be2e2e8d2b0":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/9"}}},"commit":{"parents":[{"commit":"850a42bed6654a624728f78c9a9547de8f1cf6d4","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/850a42bed6654a624728f78c9a9547de8f1cf6d4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:47:28.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a995a2b21e638235574c002d5fa90be2e2e8d2b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a995a2b21e638235574c002d5fa90be2e2e8d2b0"}]},"branch":"refs/heads/master"},"66008dd93ce0105089edd537b4b1d82f66b3d688":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/916467/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/916467/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/916467/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/916467/10"}}},"commit":{"parents":[{"commit":"96551bf7a72898aed02b003fa63a39ffc0714070","subject":"api: Remove errant actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96551bf7a72898aed02b003fa63a39ffc0714070"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 23:16:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:53:11.000000000","tz":60},"subject":"api: Keep track of extension controllers","message":"api: Keep track of extension controllers\n\nWe need to be able to resolve the original, unversioned controllers, so\nthat we can retrieve additional attributes of those controller (namely,\nthe \u0027versioned_methods\u0027 attributes, which contains references to the\noriginal versioned methods rather than the \u0027version_select\u0027 magic\nmethod). Register these things slightly differently. It would likely be\nbetter to fold these action controllers into the main controllers, but\nthat\u0027s a lot of code motion that I don\u0027t really want to do right now.\n\nChange-Id: I174c3bae39625b74e4b416e3559eb79ad8d34948\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66008dd93ce0105089edd537b4b1d82f66b3d688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66008dd93ce0105089edd537b4b1d82f66b3d688"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
