)]}'
{"id":"openstack%2Fcinder~916468","triplet_id":"openstack%2Fcinder~master~I0c8b299d01262a7b039e3d74f8f7464d80c87bb6","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-12 19:03:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c8b299d01262a7b039e3d74f8f7464d80c87bb6","subject":"api: Remove AdminController controller","status":"MERGED","created":"2024-04-19 14:51:42.000000000","updated":"2025-11-13 03:28:21.000000000","submitted":"2025-11-12 19:03:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916468-openapi","meta_rev_id":"85bc0d4306725b7436abbe3083e13874f6431e72","_number":916468,"virtual_id_number":916468,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-12 19:03:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-11-12 17:01:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-10-30 14:59:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-11-12 17:01:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 15:18:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 19:36:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-20 04:51:14.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-20 06:02:12.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-20 06:54:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-20 10:15:10.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-04-21 02:51:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-21 12:32:26.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-23 10:39:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-24 12:02:09.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 15:42:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-25 16:09:14.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-25 18:03:49.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-04-25 20:37:58.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-05 16:57:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-11 04:20:52.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-27 12:20:27.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-09-18 14:15:54.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 20:18:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-19 01:59:09.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-19 18:36:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-21 23:15:36.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-02-22 01:38:16.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-02-22 01:38:29.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-02-22 01:39:09.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-02-22 01:39:17.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-10-24 14:51:35.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-30 14:59:34.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-11-12 17:01:50.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-11-12 17:02:08.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-11-13 03:28:21.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"8b081b31ab46e1f9e3b22b47b555eb794e313fe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 14:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23175af81fcbeb7f18eaad7b8315cb66c0458c96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:18:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c62e5294baa94171a2147a17a567282b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0892e19cf284874a08ccf1bf7693f82 : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e8522b8d0bb4eb8a7ff29d8e28feb31 : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f10063b06740ebd79bf4c441aceafc06dc2dda75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-19 17:27:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"25bf44e8d41aa2518edcde8627bc6d6dbf6efce2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:43:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a48a9b4e0254d168fe9d3fc26c84b0a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da264e9b6f694d59a2bfbc21fca1ed97 : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30350e4054c84904afa6b4a6c0d71c09 : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bcc423f6a7c0c415d56b724857ebb73639091c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 19:36:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec44e666379b4ca98fa4ed56789092bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1007e77aaba4a32a75a76514d4ffcd3 : FAILURE in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd30ba681d78436db9ff90fcfaa9727d : SUCCESS in 11m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0001c7c695eb4b55915a3246eae1919e : SUCCESS in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7dafcd58402c4635898c4cfa91dfb6eb : FAILURE in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d31493438bc94a94a74f5c3632e662fd : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/d4e39cca5e574896aa0888ea0a86e47e : SUCCESS in 56m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6b6f303c4f54078b6e742b95d010595 : SUCCESS in 1h 33m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d139461bce94aacaffe89081a6e1a2e : SUCCESS in 1h 05m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e9a17ff526fc4fa7b177b966e1aff7d8 : SUCCESS in 12m 36s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f45ba07ef60a44459217f5a65aaeb596 : SUCCESS in 12m 55s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a8b073ce93024139b6d5ed6597a2535b : SUCCESS in 6m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a4ddcc0d1d64142a5aa3cc7d32215bd : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e6c538bc51542c7a313303ef30509ef : FAILURE in 6m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3ca0b2709561450584c9f772a9978ac6 : FAILURE in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/66aa4a040c9e4068ac0d0ec50d0f78c5 : SUCCESS in 45m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ae63c7d78514fc9a87b964c0637c257 : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6945c06ccca0417fafb637a4b00b3001 : SUCCESS in 1h 43m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/afb4685a9d5e4e2aa403a6304f3bc6e3 : FAILURE in 1h 41m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a9a7c0633d2246b48e62157333852a75 : SUCCESS in 1h 58m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ed65457090b842e6b47af0eefc614825 : SUCCESS in 2h 01m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2766d3172f3046e8872eeb2a2a4ac117 : SUCCESS in 41m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5f2d58d6fea245079ea0baf53a55e695 : SUCCESS in 1h 05m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/704588b64c844c9d98cb59e6cc10f215 : SUCCESS in 1h 34m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca287425ebba4c48ae922ec70a4bebcd : SUCCESS in 52m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/49e86540f0784015a956b28088709bc4 : SUCCESS in 1h 18m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e32fbadaf7f24a43bbc6ddc4175d2aaa : SUCCESS in 1h 31m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89f2a78ba2984d739fcc4061723e8d79 : SUCCESS in 1h 25m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d81574ad0ee4dd9802d7b2a4faba2a7 : SUCCESS in 44m 50s\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/api/v3/test_snapshot_metadata.py\n  Comments left for invalid file cinder/tests/unit/api/v3/test_volumes.py","accounts_in_message":[],"_revision_number":2},{"id":"a3afb2ccd42f22b5cb679a527eff4a85d1a7f11c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-20 04:51:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/68/916468/2/ : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"72ff39f68a43e0d924d5df7e3ce8623e790414a3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-20 06:02:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916468 : FAILURE in 37m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7a3ccaa963fc509dd1f09dbb09e2ee18d9a9d45c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-20 06:54:58.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/06a4b26ace5245df94a83330f8049f88\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/941efeeeaf4e456bab391a94cabd30ca : FAILURE in 1h 30m 32s","accounts_in_message":[],"_revision_number":2},{"id":"753d8695380c847d575b1dc12da1004f4f327d5d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-20 10:15:10.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/2/fc : SUCCESS in 0h 40m 03s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/2/iscsi : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b17cbc893eb782d165c92314e304221e5dbafe8c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-21 02:51:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/916468/2/check/DellEMC_Unity_FC/0a2ea7c/DellEMC_Unity_FC/1721 : FAILURE in 1h 36m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/916468/2/check/DellEMC_Unity_iSCSI/0120a78/DellEMC_Unity_iSCSI/1751 : FAILURE in 1h 39m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"dd8df265bc9bc31bf2e37a8d582fb85957a4d7f0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-21 12:32:26.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-916468-1/FC/ : FAILURE in 28h 9m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-916468-1/iSCSI/ : SUCCESS in 24h 20m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"127a8511e91feb8c8a39ee81850e3e87087847ce","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-21 15:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/1/fc : SUCCESS in 0h 40m 30s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/1/iscsi : SUCCESS in 0h 49m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f56c70f5150cb39041061b12f1b9d577ec436de4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-22 00:49:01.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-916468-2/FC/ : FAILURE in 30h 46m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-916468-2/iSCSI/ : FAILURE in 28h 11m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"e2293c486814423824f859b8162fcfebd5f3396e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-23 10:39:19.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/262/.","accounts_in_message":[],"_revision_number":2},{"id":"0dab6b76d0aa57536e7e45b3a9d7734955bc315b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a5586f32d3d73ea6331149d6492bd0ce34ffdd5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:22:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd54d5ee6f5a4ce1af937083ab21a610\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5a0deb357114a48a6a881f000893abc : SUCCESS in 14m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba89cb28b3d744d3abb968d8870f4693 : SUCCESS in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dc63ce937b094561b6324a3e271aecd04543d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:26:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a93f7088b1c341b8a865bc35f58b7189\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e208ab6cd4141c9a68e27ff075d2337 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2006b1cd9b4401b9f21d405a244cb7e : SUCCESS in 11m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c14a29d812314888a558d7c2a19780e2 : SUCCESS in 11m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61a3b130511b4235b9b1194c4788853a : FAILURE in 10m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618e06e82a2d463daea874882532c56d : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/c459523b4952442699d36df67d47d31c : SUCCESS in 1h 00m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74579f6279b146fb8b16049f404058b7 : SUCCESS in 1h 14m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/003d51925d404ecb99869047c31520ff : SUCCESS in 1h 09m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fc7bc4611811483394289869d44515ba : FAILURE in 13m 26s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6aea18e0395a4daab1776e3eff699695 : SUCCESS in 13m 21s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b37ea42051394b7a9a8ee57536740f66 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/249f0d6346024092bfb514bc39a50567 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/86e7f4ee56c444d8811f8177459cfdad : FAILURE in 7m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4b774547adda4dacb13bc8b052dc29b6 : FAILURE in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25f5b7a22ef9442d803a7e82be720443 : SUCCESS in 45m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/04fb2d974c834d5e8ea745cc87ea8f6b : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/93dfdad586614d0e8ddf78f9b1e34829 : SUCCESS in 2h 07m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/296294aeb659405a80aa3caec65ab7e1 : FAILURE in 2h 20m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/736f9d8b968b487bb8d1c4dc261bd2c9 : SUCCESS in 1h 54m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6e35716a5fea468a9d6a5e768e0bfff3 : SUCCESS in 1h 39m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/732f81c46de7401cb7a8f799a1aa8458 : SUCCESS in 39m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/326696be88244fd0acc9c2d011b98ef0 : SUCCESS in 1h 10m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dd449f29a54c4d8b9568cf31e6e2f030 : SUCCESS in 1h 39m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/015e92a6f3f143d3829e22ad1b46d6fd : SUCCESS in 44m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6fb329c806634f6a978f7aa520e8e11d : SUCCESS in 1h 21m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2f6f65f273d840b58a1c75e2c9b31849 : SUCCESS in 1h 40m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/128eb254a9fc40338f9bc235dce80746 : SUCCESS in 1h 28m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/137afbdf5a19423a815bece70174d98f : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":3},{"id":"bcff048708a02cd72d9c6e58d231fe25202625ca","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 16:04:19.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/813a805d82cf4f7facdeb6a4761919f5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/391248dd9b9943a49cd4c2f6c4b4dbc9 : FAILURE in 20m 46s","accounts_in_message":[],"_revision_number":3},{"id":"59e16280dedd23a0294b82c0d8327a2025872273","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 17:45:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916468 : FAILURE in 12m 49s","accounts_in_message":[],"_revision_number":3},{"id":"1a5cb80ce49c5e9d5d0318660d73907d5e10bbcf","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-23 19:00:06.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/3/fc : SUCCESS in 0h 48m 52s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/3/iscsi : SUCCESS in 0h 53m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ad36d60a8ee418ea681e0af4c2c1b91b3be90f43","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-24 12:02:09.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/68/916468/3/check/fujitsu-eternus-dx-iscsi/80c981c : SUCCESS in 3h 13m 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/68/916468/3/check/fujitsu-eternus-dx-fc/2849755 : SUCCESS in 3h 19m 23s","accounts_in_message":[],"_revision_number":3},{"id":"cdbeed4d0f7182839e811a944f952b03b520b252","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-24 19:28:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/916468/3/check/DellEMC_Unity_FC/833014d/DellEMC_Unity_FC/1755 : FAILURE in 1h 30m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/916468/3/check/DellEMC_Unity_iSCSI/eab2d6f/DellEMC_Unity_iSCSI/1785 : FAILURE in 1h 55m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e0f6c9cc2986edc90c3b10167b6b7b9238677aad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-25 10:54:05.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/296/.","accounts_in_message":[],"_revision_number":3},{"id":"945560a8a3c01fce3bc418e2d677faf6e8e28734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-25 13:02:12.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1a634aa228f8cd4eeb74fe6b957e45e419bf19ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:21:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e0052c3daf24a86898041dd087db18e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/224675f75ece42e6b181daa1a7048fc4 : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0bf123337bde4c158b81cc548d21bac0 : SUCCESS in 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c1db5b4198c57deb6967ad8eaba272a2d2b336d3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-25 13:48:06.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-916468-3/FC/ : FAILURE in 35h 51m 55s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-916468-3/iSCSI/ : FAILURE in 36h 16m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"c918a028e9bf74ca4fcb966b287118c5fbe6ebfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:31:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03aa42194a434bd0a1a9b6cafdad435b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94a59a7747b5487f9f4376ff3790aac6 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a3aefd8e37f40dd829746b746678d6c : SUCCESS in 11m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7696b1716f04be984f3934b16436d68 : SUCCESS in 11m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80e0a0eb7ccf4ac39d6116219ca33251 : FAILURE in 12m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9b1a605dda4f6a9456515138cd72a4 : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/f08657454fe84765a58cc71724c809d3 : SUCCESS in 1h 06m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f631418e412043e6aff325abcbcb683d : SUCCESS in 1h 35m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54c6c55f231441c48009790d3fddea73 : SUCCESS in 1h 00m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d6519d360fc4f9184e8de0d0cefa4d7 : SUCCESS in 17m 51s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a6daf85ae67746788816f8dbcfcd9e9c : SUCCESS in 13m 30s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/74d5dd846ee14d7789cd6f8aed564594 : SUCCESS in 6m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/471f4541d5b14b9e9819a8eb30f1395f : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9eee0d8918614e35a01a5a50df9b8fad : SUCCESS in 6m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6a22dee6047a4b4e97e3bea5681a7657 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1eca06d62b8a44a1a62699af42f87c6b : SUCCESS in 39m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d831ab01957f4943b6a8410f13200bf0 : SUCCESS in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a9dc12884d24e5990c3a40d8d7ddec7 : SUCCESS in 1h 59m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d9b1cbe9fbd9451790c412b43d8e5abb : FAILURE in 2h 21m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c38b21361cd14905bbd99713577de133 : SUCCESS in 1h 46m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/008f0b448a59482d857a9751078e702e : SUCCESS in 1h 28m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0759b6b3c0174af8b38cc402828a6e4a : SUCCESS in 41m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0f5ed8b84c9a46688488fa665ba6e41c : SUCCESS in 1h 17m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/10dc28b923234e05a3cbbce044148fd9 : SUCCESS in 1h 39m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5abc66cf0d694dbb832d1a4d06880308 : SUCCESS in 46m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/480583ce1cba48cda3a213e39ce05c3c : SUCCESS in 1h 19m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/db2aecaf413b473699d287283eb2d7f5 : SUCCESS in 1h 41m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf86eef648404d2bafd2c73d49421273 : SUCCESS in 1h 22m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d59efc2f6b214d4aad4845d6f857fc64 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":4},{"id":"a03604443f502d6c583ba041e6af2d24efc57ea2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-25 15:42:49.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_04_2024_15_36_45\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"bb2b125ca14f77426e70f4be5a95af5098bc0cfd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-25 16:09:14.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/916468/4/upstream-check/cinder-tempest-plugin-ontap-fcp/b266638/ : RETRY_LIMIT in 4m 36s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/916468/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/b5a3027/ : RETRY_LIMIT in 4m 03s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/916468/4/upstream-check/cinder-tempest-plugin-ontap-nfs/fb9d4a6/ : RETRY_LIMIT in 4m 12s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/916468/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4c3cf03/ : RETRY_LIMIT in 3m 52s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"5bc310082ce8bbbb978b1e4b32bf60560207c0e7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 18:03:49.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916468 : FAILURE in 24m 48s","accounts_in_message":[],"_revision_number":4},{"id":"43d7faf902f40ab39c32f03adda174ea5951007d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-25 18:34:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916468 : FAILURE in 25m 00s","accounts_in_message":[],"_revision_number":4},{"id":"0b32ec64b6cfc608997c1a69a91bc7af23d631c4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-25 18:52:50.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7d458389c6904bf4af06fbb232d05602\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c7cf13d9603e4c8e97f65f558a122111 : FAILURE in 1h 32m 28s","accounts_in_message":[],"_revision_number":4},{"id":"8b173948ccec1c55bb3adf578e5763bb6ec4b5c2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-25 20:37:58.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_PowerStore_iSCSI/d7992bc/DellEMC_PowerStore_iSCSI/1276 : SUCCESS in 1h 37m 26s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_PowerStore_FC/32c8a64/DellEMC_PowerStore_FC/1447 : SUCCESS in 1h 02m 17s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_PowerStore_NFS/ee9e942/DellEMC_PowerStore_NFS/966 : SUCCESS in 1h 08m 59s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_PowerStore_NVMeTCP/72d7dec/DellEMC_PowerStore_NVMeTCP/827 : SUCCESS in 1h 56m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"5f856ff4ade7951807dbfc4f229508ff342de1c0","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-25 22:57:14.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/68/916468/4/check/fujitsu-eternus-dx-iscsi/393798e : SUCCESS in 3h 06m 35s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/68/916468/4/check/fujitsu-eternus-dx-fc/e69f9ea : SUCCESS in 3h 05m 53s","accounts_in_message":[],"_revision_number":4},{"id":"045c03d550ce0a1f7a8f4d0650b0fc71a6c39c33","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-26 07:39:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/311/.","accounts_in_message":[],"_revision_number":4},{"id":"eb3a67f75bc1251804128dfd7f9d91a95617a913","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-26 08:54:41.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_Unity_FC/10e0076/DellEMC_Unity_FC/1776 : FAILURE in 1h 32m 05s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_Unity_iSCSI/4260787/DellEMC_Unity_iSCSI/1805 : FAILURE in 1h 55m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"afb6d081391fcba9147c41a7efab71f461f86736","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-26 10:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/4/fc : SUCCESS in 0h 54m 33s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/4/iscsi : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":4},{"id":"b318863252b17fc4861d47e4099a27bec0196b70","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-29 10:45:45.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"3fdc1e343ebd947567922ec764d7267e6b64718d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-05 16:57:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/916468/4/20240505211332/NEC-ISCSI : SUCCESS in 4h 19m 14s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/916468/4/20240505211332/NEC-FC : SUCCESS in 4h 44m 21s","accounts_in_message":[],"_revision_number":4},{"id":"4654e900ad339cbb2dd1e509027ff14cb661a38a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-11 04:20:52.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_PowerFlex_3.6/5ec4179/DellEMC_PowerFlex_3.6/1276 : SUCCESS in 1h 37m 19s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_PowerFlex_4.5/1a2663c/DellEMC_PowerFlex_4.5/210 : SUCCESS in 1h 50m 25s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/68/916468/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/b357821/DellEMC_PowerFlex_4.x_NVMeTCP/151 : FAILURE in 2h 54m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"d813e319005be01016abb19534b1a370599d4c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ce255adeaf3af1f56ae2ddd2c4ccdb409c669902","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:20:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2588e2dde5747dea83942038e711f8c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/005f64284af14146896fd1072a765ceb : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f0f9869ea6840f1bcf42c27f4631062 : SUCCESS in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"23b54ccc14762439dfd79df5b20bf54f2f909e7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:38:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84c16a6d597c4d1683a5a7473a4782df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183990aedc624531a0a15a0065d1f942 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a437d69e289541ccba5138939fbacd12 : SUCCESS in 11m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1a5c868d28b4d3abff377509d890a2c : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e2af0def1bb4ca79c130ed3e8a94183 : FAILURE in 13m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41b1997e69f4ee1b1bdb2ab42aee7c3 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/1c053087cb5d4c809953a397ec75bf21 : SUCCESS in 1h 01m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2238aede8b294cfd8deedf6ee468a2ea : SUCCESS in 1h 22m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ea2ed62b591443181cca241d6357c5b : SUCCESS in 59m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d0911d8472f14fefb8ca957643928835 : SUCCESS in 18m 33s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/18ed40934cad42588a17a77b2ae66e71 : SUCCESS in 15m 54s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f6c528ded174497882bb7fdb8e0a9da : SUCCESS in 6m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c8a264de33b4a8c8125a596cfa30fda : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f475344426c149d8bb5a96b33568ecce : SUCCESS in 6m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/14ce3e99d4c743939e2b479f87653e64 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/04a97a6b36ae4d07967ef38089d3cc51 : SUCCESS in 47m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/299982b0d5534f29bf08e6ea6f998a78 : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54b8845dbb164a64a839b9c6f931a4c8 : SUCCESS in 1h 44m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d14e6138cd374cea87ce7c6d93dd7803 : FAILURE in 2h 03m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/728b1e1fa6b9480ba4ce4652e6e4911e : FAILURE in 2h 11m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/77f70c72a3fa477aa27be8d75146d6ab : SUCCESS in 1h 44m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ce85da2219ce483d881e8891b11cda3f : SUCCESS in 41m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/904aa0162ff44363a0fc48c74ee015ce : SUCCESS in 1h 21m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f07b051864444b56aa444660b325708c : SUCCESS in 1h 30m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c2473a31654a46bca74b1b3c75084f4f : SUCCESS in 51m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dc3af729e2494e35b3f41339cc9c9e8b : SUCCESS in 1h 32m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a0acdbcc377845de8ccfd3b075df4ac9 : SUCCESS in 1h 04m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d6ecfe214a354b06b182fb43baf3db08 : SUCCESS in 1h 23m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddc194dcda144dd3957d5ef7a47ad923 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":5},{"id":"5e2e6ede68c05fa375328b11d160d09e43ee15ec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-14 03:41:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916468 : FAILURE in 47m 15s","accounts_in_message":[],"_revision_number":5},{"id":"80ddd5b86d976a0d9afc3eb370151ef56d5ca3e0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-14 05:12:46.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/829cafd81dd747aa9b7fb5b99d88a1d3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f757b1110c544cf2afcb3de6f7930304 : FAILURE in 1h 22m 42s","accounts_in_message":[],"_revision_number":5},{"id":"62fee85101026d8586bd404cc8b4c093c32222a8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-14 17:20:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/379/.","accounts_in_message":[],"_revision_number":5},{"id":"07ac5084867d156320dee368c987fcbaa859392a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-15 02:48:06.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/916468/5/check/DellEMC_Unity_FC/ee2a23c/DellEMC_Unity_FC/1858 : FAILURE in 1h 47m 44s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/916468/5/check/DellEMC_Unity_iSCSI/ce9213a/DellEMC_Unity_iSCSI/1886 : FAILURE in 2h 04m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"c461709db4edb000e6b59e8a49e6f00812a8edc2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 08:21:44.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/68/916468/5/check/DellEMC_PowerFlex_3.6/2971018/DellEMC_PowerFlex_3.6/1340 : SUCCESS in 1h 33m 33s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/68/916468/5/check/DellEMC_PowerFlex_4.5/78384e7/DellEMC_PowerFlex_4.5/265 : FAILURE in 57s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/68/916468/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/bccc9ca/DellEMC_PowerFlex_4.x_NVMeTCP/212 : FAILURE in 1m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"41127486580c690bf4bad3b848cb8b5ad3115bd0","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-15 13:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/5/fc : SUCCESS in 0h 41m 00s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/916468/5/iscsi : SUCCESS in 0h 51m 27s","accounts_in_message":[],"_revision_number":5},{"id":"da2c20f3d0718f8c2a2a99f4ad3c98fcd31cfbdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"9c4cad95aafbdc6128389b99641ca68eb0651dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"d81a148460054b2e0549ccaf1a3ae6a9f9e8bd96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:47:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff62384831504d2e9f962857acd4b4af\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c333706000eb43e88714fd98a68c493b : SUCCESS in 20m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/524bf063dd8e4c478c7b572735ec4859 : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5371ac63732ef7a11d90dab9c0d459edb345200b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:53:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/494b8ddcba4849f5b61b18edd2196733\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef5266e357941efb6420c2c6b49140f : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e535af1af154df7adb016b32bf39b8f : SUCCESS in 11m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15306cd504c74825b51a1d2816c28f78 : SUCCESS in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71812155fb2344d6b78a30cdc12c8685 : FAILURE in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dccbd9fc5154deda2dc51f00c0e37bf : SUCCESS in 17m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/8cca0122a08f4956b0e876556ce7cc69 : SUCCESS in 1h 16m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a918b816c2640c6afd7f7b6ca11297d : SUCCESS in 1h 20m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35af3736e9a74fd7a610cc093cb2c67c : SUCCESS in 1h 01m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/589b9f1f7023404bb829b83eef637649 : SUCCESS in 15m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7e19c63ee24442688f94cda842a5600d : SUCCESS in 6m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83b131d2684e496db66e2419d091b4b4 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b1c6d4c4df64ce6afd007d12c91e4c4 : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4265d31314e54afe92fb87290e83660e : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/203d619e46d643649742c6a39926ca3c : FAILURE in 38m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1985f95ac7334300a99fdc988477c890 : FAILURE in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e24ef5b0311943e0ae6436cfd158e6c7 : SUCCESS in 2h 00m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ef30f72eca444555bce1e6ceda428045 : FAILURE in 2h 37m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e240701785d44945b508613a5db1eef0 : SUCCESS in 1h 52m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/de3bd22bcd114789a14851be82b7fef0 : SUCCESS in 1h 53m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e301623c210e434eabe8d3189e88b3bb : SUCCESS in 47m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b9f3391e2ce8472a9652d813008d0b55 : SUCCESS in 1h 19m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c9401469efb14c1b8f5275c107f8cb35 : SUCCESS in 1h 51m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fce5b1d18f1745a89cfdf17b2718c3e7 : SUCCESS in 53m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5f3ff4a6acf54b11944c305a6e8addb5 : SUCCESS in 1h 28m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2f11f0841ed44d7f9e3e522fa945cb55 : SUCCESS in 1h 34m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57fe3603b110442bb7f121a7f3cf2007 : FAILURE in 1h 47m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74ec48f8bef64f4c8bc39d1c38373281 : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":7},{"id":"c438b1e526139b6818a203817ef36737e926307c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 16:05:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"03d3aaf751c7f9c9b8d2219ee7aa942effba9153","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 16:19:47.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e861da905809416189aef7f4f1c7555e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c20b23bf9b054b9f8c34e44c811bd5c5 : POST_FAILURE in 22m 03s","accounts_in_message":[],"_revision_number":7},{"id":"0db6c7d812ff056ef69b997e01e5a70b22ce33c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:27:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ec37978d57462eb084e6f6b49563d1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9bae6a3d23b84789ba43d6649aaf82fd : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6b1dd3e1d06543d7845d5cfb5ea75222 : SUCCESS in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"98bbdda1965a8023ae49bbcac5ae5ca0e022d4bc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 18:17:40.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916468 : FAILURE in 23m 50s","accounts_in_message":[],"_revision_number":7},{"id":"0f1645ffd974dc7b8ddc56b6b75d58e016981957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:24:18.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95f9ec227827401aa2e2eb11b14257c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe1749d4edf64d7a968e4fb443a7da34 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b91c2c7a78b477f85daa393e13743bb : SUCCESS in 12m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b991117934624792865af3b3fe568444 : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3fb59d2d15a4b9b88cb0379a0fc48a3 : FAILURE in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc3d11ba65b6435699c5c34a0e07afd7 : SUCCESS in 14m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/bd411864bfba4b208ffe10ef88ee38a9 : SUCCESS in 1h 04m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e9900f50fa2460985a2f71782a39311 : SUCCESS in 1h 32m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64af585f0a904ef7a7aa4141798e01c5 : SUCCESS in 1h 06m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/762de671e9174a44ac53b8dbfa1d7a62 : SUCCESS in 18m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f2b20cdb2a84d28b1e408a54330c2ea : SUCCESS in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d0f2229b34b4c5d9371ac9aff313e88 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95b87b6ac28647e29ed5cf885e4da756 : SUCCESS in 7m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/266f7ac52d2e4812b4afbce418429435 : SUCCESS in 8m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a9724d1266eb48158d4e0e8c95c43600 : FAILURE in 33m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6678308f750e4918815e447e3d775b76 : FAILURE in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c533cc22300c4bbab6491c222854b13e : SUCCESS in 2h 02m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2a43d2e336c640168c39483891adea18 : SUCCESS in 2h 01m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f6e4e61f553468aba68f763fc71d1be : SUCCESS in 1h 43m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2bf4ac12d2ee479a870719480e0f6df2 : SUCCESS in 1h 53m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/004123b20cdb4d0d86b28b88ef5d6982 : SUCCESS in 37m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b640e4d23e64445085993cff72c7a68e : SUCCESS in 1h 07m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/26f19e0784d24604b9cf2daa5252d543 : SUCCESS in 1h 24m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/593a56dd4b7944a89ca4110ec4bdeee6 : SUCCESS in 32m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a9302cd9f47e4dd3bed676e76e09c53b : SUCCESS in 1h 30m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/170080f1fbaa42aa83b657d1c29e7ef2 : SUCCESS in 1h 27m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5332930836cb4dca8f6d93cbda31abcd : FAILURE in 1h 34m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89c7146d0ad2401eb93c0a1ce9e72052 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":7},{"id":"934474573ba7e9ddc653a971825f0c394b00620b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 23:31:33.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916468 : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":7},{"id":"73a1fed1b0d129a85ed8cd6456b411259421dfe4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-21 14:23:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/441/.","accounts_in_message":[],"_revision_number":7},{"id":"6a60cd7e0ec906a716e818b77a89abd54f885662","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-22 03:25:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/916468/7/check/DellEMC_Unity_FC/3e3fedf/DellEMC_Unity_FC/1918 : FAILURE in 1h 57m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/916468/7/check/DellEMC_Unity_iSCSI/7829045/DellEMC_Unity_iSCSI/1946 : FAILURE in 2h 06m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"2ffeedb43384c878ba5e0fd74dfbff6f5f0da592","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-25 11:18:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/916468/7/20240525120500/NEC-ISCSI : SUCCESS in 4h 52m 33s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/916468/7/20240525120500/NEC-FC : SUCCESS in 8h 13m 22s","accounts_in_message":[],"_revision_number":7},{"id":"15eca44823284fed47eefd5ee4a0b085f1dd9000","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-27 12:20:27.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/916468/7/20240527132618/NEC-V-FC : FAILURE in 7h 51m 20s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/916468/7/20240527132618/NEC-V-ISCSI : FAILURE in 7h 54m 05s","accounts_in_message":[],"_revision_number":7},{"id":"d2b48769e685ba435a75f69898ad96f59ed7e79f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"01126efd217002c71ac94ad5c1a0e1445df60596","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:54.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/596/","accounts_in_message":[],"_revision_number":8},{"id":"58bbae4d4dbd5f5a0aa7adf128956395c1c14a53","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:56.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916468/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916468/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"71c1d7260dbf55177ac2b7d166edc5f8d18a9517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 17:44:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/665bfae304e74004987d5dc4a9dd9b12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7038226357fa4a0ba46e4233bb55ac36 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fff62ec245f64ff792e66ef4b9902602 : SUCCESS in 12m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a6055cef38e46e4b40b5d5548b60d35 : SUCCESS in 10m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b8f94744f0c4cafac5fcd2c14cfebf8 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a44b4430c14e73a21cf1f5cebe9e63 : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f9cbf7dd6b764be0a74b807a115d1ca8 : SUCCESS in 1h 05m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/107447d47388428ebff5e44bd486f33e : FAILURE in 1h 39m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b633c1bdf4f43ffb513df88e5ad2c37 : SUCCESS in 22m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ddb5d2b084d454c95ca5384d2a1c89b : FAILURE in 19m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fea5637cfc2d4359b300ccca87639bb1 : SUCCESS in 8m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/287e8bba64a64abcac3dd7f891e4f4ce : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/77f1f688926d46118ac9333a9e715c6f : SUCCESS in 6m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/931e112d53f1408a875774d64590928f : SUCCESS in 9m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82e679e964ea49bdb19267d6b8471de7 : FAILURE in 36m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e36bf4ffb9a342049dce6108bc9f1089 : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3037b2ab32f45f190cfd5e057170152 : SUCCESS in 1h 49m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/366e2c10cc5e4787bc2cb12faee96dd9 : FAILURE in 2h 38m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14d3ba5f00c041f295c71559172db6c4 : SUCCESS in 1h 57m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5a76e24f072d481fb5a2b36e5175f9cc : SUCCESS in 1h 53m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1654a3c2a9cb43eaa701653fc1a3720c : SUCCESS in 34m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/323063f45907472bb6995ea50e23fb05 : SUCCESS in 1h 30m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d0a0af4333a84216ba9c2d96ea76a704 : SUCCESS in 1h 37m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/57623cafa2384eec8e48543b8e057cd9 : SUCCESS in 41m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6fd67dfdd15c4661806664406f1482f7 : SUCCESS in 1h 39m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2b2fd5f7e3974c72930dc9f4202c3caa : POST_FAILURE in 2h 10m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48b8e9ce61d246c4907f81445e277de4 : SUCCESS in 1h 28m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1414185540a843b79b362260b3983482 : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":8},{"id":"b6d523fce5a073460000912dc23e6a30582fa72e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 20:18:19.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d29f51e59271448a8c0e649bec4902a7\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7aadab532ef04a7ebc4badebf1f41a83 : SUCCESS in 41m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9a1ac55fb043489cb89f1ffa0149b46c : FAILURE in 43m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5fee07034d424843b6d2671ac21d543c : FAILURE in 1h 36m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1fe1b4432aa64dbc992775fe24b4af2f : SUCCESS in 37m 36s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"20feb101d0fd2d4d13b02eb060af9e2c3750f7f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 21:24:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15578ee33bc44620b30331dfd3b244a9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58ad154bb6a34e088ce46d9dab794e7b : FAILURE in 23m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7950454f5d24820b3cd65ba8ba726da : SUCCESS in 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"521f0e9e411ab411c6d1215e104be3660d9bfc13","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-19 01:59:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916468 : SUCCESS in 43m 19s","accounts_in_message":[],"_revision_number":8},{"id":"4a2d0f1b489f0a40837ed0db524a55459692aedd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-19 07:45:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916468 : FAILURE in 19m 18s","accounts_in_message":[],"_revision_number":8},{"id":"2b3341b1833acd52953d1de8c454742498f428e5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-19 07:58:17.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916468 : FAILURE in 47m 47s","accounts_in_message":[],"_revision_number":8},{"id":"d29bde56a24c814f62266fbbb989496989e88371","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-19 18:36:45.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b4be5c55d4e44bee9d4965f6049b86e8 : SUCCESS in 1h 14m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6b7fbe48438c468683adb6de1a769206 : FAILURE in 1h 19m 41s","accounts_in_message":[],"_revision_number":8},{"id":"c5e1e3e917133dbeccabe30f0ad89219eda02041","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 05:41:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/571/.","accounts_in_message":[],"_revision_number":8},{"id":"828d509fb37d27bf730a9ea3bb5c15bf5df9ddf3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-20 18:55:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1159/.","accounts_in_message":[],"_revision_number":8},{"id":"72632f89b93b7492a68f337896c115de6d2a6e21","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-22 05:51:03.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/916468/8/check/DellEMC_Unity_FC/07e1da4/DellEMC_Unity_FC/2638 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/916468/8/check/DellEMC_Unity_iSCSI/27c25ce/DellEMC_Unity_iSCSI/2647 : ABORTED in 6h 00m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"0d46e405a4a212de8db8679e9d83a7df071c4042","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-23 23:02:53.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/68/916468/8/check/DellEMC_PowerFlex_3.6/5d42503/DellEMC_PowerFlex_3.6/2132 : SUCCESS in 2h 19m 21s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/68/916468/8/check/DellEMC_PowerFlex_4.x/ae8d629/DellEMC_PowerFlex_4.x/1016 : SUCCESS in 1h 55m 21s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/68/916468/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/52bf290/DellEMC_PowerFlex_4.x_NVMeTCP/956 : FAILURE in 5h 54m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"da7951a1567ca98499427a012cb2ebb3046d1167","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-24 19:45:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/68/916468/8/20240925005243/NEC-V-FC : SUCCESS in 3h 48m 31s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/68/916468/8/20240925005243/NEC-V-ISCSI : SUCCESS in 3h 52m 42s","accounts_in_message":[],"_revision_number":8},{"id":"aed5550a910ebf54f4ab17d5247866040b626e19","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-21 10:53:56.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/68/916468/8/20250221182557/NEC-ISCSI : FAILURE in 52m 01s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/68/916468/8/20250221182557/NEC-FC : FAILURE in 1h 27m 57s","accounts_in_message":[],"_revision_number":8},{"id":"46c8edc73382eb4c6c35764b2deb016dd477afca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-21 23:15:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8c228df6f72a7cc7e7d2fc22f2dcaf27f263a46d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 23:46:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47fa9e6ec2364a458c6371ebe28d06b8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9579f229bbb643af9154adcf8491d83d : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37a341b8eda74e59a07eb4815c9e4c9c : SUCCESS in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f57654df07bb858a9d545c1dff31408aea491c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 01:37:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3eb4cf2a3214c34b2747b23205d3ab9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/765b68da975442c2aebd83ba8b5e838e : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/762a1268c640419989397807b9573711 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aca5a6012d5746cfaabad5454f8db504 : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a6a84600ef41779ef52426a697d2a0 : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/26ec20a6aa7647bf8024ec2ebcc739e1 : SUCCESS in 36m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b58cdf8efe464ae99dcae909280250e1 : SUCCESS in 1h 08m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d739f03d9cd4151a99f1bed4f620811 : SUCCESS in 1h 20m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3187eada7604a38a338990932063f28 : SUCCESS in 1h 03m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/930810f8e5b04b23b9361e8073e1f7c8 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1c1c966e362b42cdb84ffc4009946609 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6a34aa1b8e04c75a66c83d508f7006e : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85d30ee5480e4f83bfd563ff86182d32 : SUCCESS in 6m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bcf19967c37b4fc9916c712ae8a3e4b4 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/254021d3051d447d9d59737a0b9dab3c : FAILURE in 34m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0b8fd520abe64942a5108b659c274f32 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9a153baa1f3b4ccdafee260c54becf24 : SUCCESS in 1h 08m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9f759c350afa4b6195b27e019285818a : SUCCESS in 1h 44m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6b120b1b98b455c93f6c4d47c2a6292 : SUCCESS in 1h 53m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/88a89f8a42494c419105c9c618ac6fee : SUCCESS in 2h 15m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e40219ec62046c2a0ddac8de4a10987 : SUCCESS in 28m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4527f75d9c0f426abd315a562c33d201 : SUCCESS in 1h 08m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6cf6a7af5ff3455c8fe731fad25a0481 : SUCCESS in 1h 25m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/30c02e3e9d724b568bc502d8abb9c2be : SUCCESS in 50m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/592596ff67b643ac8c7fd1d0a97ec932 : SUCCESS in 56m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/52f9289a68be43a88743c2b5d643ed61 : SUCCESS in 1h 47m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6083baaccfcb47658483d42e7147dcfa : SUCCESS in 1h 53m 59s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96a279298c6e4f92901a8167bee14aee : SUCCESS in 1h 20m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67dbe34cb10445aba28a13775b0b1278 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":8},{"id":"dc941f9ec9a36e139c3febe2273ea75838b0a64f","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-02-22 01:38:16.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"9f7fc96030929c175f0acfcc829adf5e063f202b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-02-22 01:38:29.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"b15f16f940a1ad96c19b3b1cbd506caa1e35655a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-02-22 01:39:09.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"1e4f851841f21c3087231fe2b0931a331ae7f59a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-02-22 01:39:17.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"3a6a45678c017f37bd93483f23894cc7cc2100f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-22 01:42:51.000000000","message":"Patch Set 8:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/22_02_2025_01_37_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"5da7907b4dc9463dcd8878c8e055018f403a7ff2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-22 02:31:57.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ab2cb8a5e73643c1b9efe089cc8e5217 : SUCCESS in 1h 09m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e11d266706d242a78248a31fcd51412e : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":8},{"id":"c7d2d238b167ec2e2de723f0447ef2510a887826","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-27 14:04:19.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"8d6010ca673d11958dfef0003605b8ba0438a5f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"f24e9ba318b0f366b7bea9f6a328553284e17990","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"d8ffc7ec01eb671342a68f802aeaf19af1a582a8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:53:59.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"02be7cde342aee2312dfb6f0ff64280a726f041d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 12:54:06.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"be0494e11d624c363a74256c5efa3e6054ff0ebe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 12:54:57.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"35ae90b521bd4f83d92ab7e592299cb74c5daff4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:21:37.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"c0faa2567c548ced07ebf6003137b45b4b85ea39","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 14:51:35.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0c8b299d01262a7b039e3d74f8f7464d80c87bb6 .","accounts_in_message":[],"_revision_number":9},{"id":"18f7d96626727427146a74aeae956642c9ebc863","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 15:20:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd860294771342d88900b077cce7f97f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc8902bd51784e239793ed394f135752 : SUCCESS in 20m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7a116787275f4568ab9b67376420f010 : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/feb2684e90a6416ca9c8a2e143dae7b9 : SUCCESS in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d623efc58c87d05f1d8a0e104a22d4976e12383c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 15:39:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19214d8ea1ba45e797c9d44db652a387\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bca475c658248b387b7a6a1f1d59094 : SUCCESS in 8m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83a275c4bc4247f1b42682fbc0442344 : SUCCESS in 11m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc781a214c444bd482ccb623dca3f9fc : SUCCESS in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97042e80ae6c43bbae43871f48b5527c : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/a0b4c0d847474070add5c52dce80421d : SUCCESS in 55m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2cbcf8907614f1d92aa11e9f9402f50 : SUCCESS in 58m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77d3ca6e14e54043b00d9495108c445a : SUCCESS in 1h 26m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0404aee6e0447809d919bce92797385 : SUCCESS in 55m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/49ffee768b794e81ab0c4c6eab92889a : SUCCESS in 26m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e8b88ffc9d4943638b702bf3c6a0fc4d : SUCCESS in 6m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6523684485f44218f0f63d13b9c9c40 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/996297bda21e420dbafecd8286f7344e : SUCCESS in 6m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dcdd91af1b0c4c3d8626addf6ac8476e : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fe1063098a2f4d9d8a6897670ca6d2c4 : FAILURE in 36m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/571c4b694b41467a93a713053446da2a : SUCCESS in 4m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/40032b5dd8f24e529d08e72daf8a4a34 : SUCCESS in 1h 40m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/49b7579eecb54c6199d221978e88037e : FAILURE in 2h 36m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2c6baed0499a4b08b044ced700feaeb3 : SUCCESS in 1h 36m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6ad1bb577900465dad7eeeef67e53c34 : SUCCESS in 2h 07m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9c61116e3d5a4aab8df9fc2bffa27f9a : SUCCESS in 29m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dbbf83dcc4404a56b2171df8d6272ffb : SUCCESS in 36m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/728d8631ea3647ed8f3e0ce73e0eba49 : SUCCESS in 1h 33m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2727a54d10f649b88d622dfd458fc1dd : SUCCESS in 28m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab636f3d334a4a2cb3a0cd6bc13a11bc : SUCCESS in 1h 35m 35s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/dca6771e874241bbba3703f453e25b68 : SUCCESS in 1h 57m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/751c21d822354db0bab19e7c9f1562e8 : SUCCESS in 1h 39m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fdf74036c614310ab85605283c8d9e9 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":10},{"id":"bcf40398c65c35422ba14f503bcbb12d256f3759","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-24 15:50:01.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_10_2025_15_43_39\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"1f602ef8ab9059ea5c9660b17aa7dc3a737d3d4c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 17:00:30.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0c8b299d01262a7b039e3d74f8f7464d80c87bb6 .","accounts_in_message":[],"_revision_number":10},{"id":"21bf70c0b8ab7ee4bbbb29763948d2feb7a9bfa8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 05:04:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2261/.","accounts_in_message":[],"_revision_number":9},{"id":"68de49db2411b2eeead14145f3a07157c3305838","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-25 05:09:01.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0cb000f99c3f48999dac1ff5be4ae103 : SUCCESS in 43m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d32bc29223454a6791ce04e4790655e3 : FAILURE in 56m 50s","accounts_in_message":[],"_revision_number":10},{"id":"761f3ceeff64feb2bc0e8f1fcc3a4813e90db7d6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 05:26:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2147/.","accounts_in_message":[],"_revision_number":9},{"id":"3563ac0cc54bde3ae83bba8a6774bbbf97972355","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-25 06:24:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916468 : FAILURE in 1h 12m 59s","accounts_in_message":[],"_revision_number":10},{"id":"ae609d7a0adc743b7dc041be33369394da30b139","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 20:33:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2279/.","accounts_in_message":[],"_revision_number":10},{"id":"6fc541fe63473e55a56e31044d58904ac8a57a9b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 21:16:53.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2165/.","accounts_in_message":[],"_revision_number":10},{"id":"3874d51bb9cbca5c5f6c902e76e88b7627dc3600","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 14:59:34.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"224db8eb52cb601a1e6a375491a5bb2cf30c3db9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-12 17:01:50.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7e65370294567adc66feae1ca8746d5d59433f2e","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-11-12 17:02:08.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/916467 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"b7e60295567aa01a3c99cb94ee22e6e7253423be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 17:02:34.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"007f06a169f39b68517c6479a473ac8b33a9c755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:03:36.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82d8910f806042a2894f7352aafff8dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c71568645a54d2a9548427b522164d5 : SUCCESS in 8m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/397179b1d71549daae33fa7f11fa7af0 : SUCCESS in 14m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d15dfd6ce784663bf1407484df25feb : SUCCESS in 18m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6735c298ef542ba8c7dd9299e2a9154 : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/115a4af5d62f40f2ab536b5160513bd7 : SUCCESS in 1h 03m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fca0a8b9401649e2800b6afdc1e573e5 : SUCCESS in 1h 05m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0fb30f141d044949a19cebf5ba0ef576 : SUCCESS in 1h 47m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c563d77ef9f4d438c68237ef4fc6174 : SUCCESS in 1h 08m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ca5b82a98f5e4458806a9e0c8776ff82 : SUCCESS in 1h 00m 21s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c194abcf5b940aab1c5d98818b240d2 : SUCCESS in 2h 00m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66efb0794960425b8d6de39ef0628ea8 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":10},{"id":"ecd32b9525381b4d3849a2d5f9d44d0c561f68f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:03:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"2dca187b6e105209113a47309477e816459cefcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:04:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a24332a69a4faa9d4fb28e47899fb1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da1669fee6cf4c68a69500c46b1da956 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":10},{"id":"85bc0d4306725b7436abbe3083e13874f6431e72","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-13 03:28:21.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-68-916468-10-fc : SUCCESS in 1h 2m 2s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-68-916468-10-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a693e055a458d4628c8bdfffbf51d7e791f29c70","revisions":{"f37aca09d0d12a0eecb3f3218b616a154c50ec2b":{"kind":"REWORK","_number":1,"created":"2024-04-19 14:51:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/1"}}},"commit":{"parents":[{"commit":"919323e30d39e9c204ba9a98fcf1324381df9564","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/919323e30d39e9c204ba9a98fcf1324381df9564"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 14:50:49.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f37aca09d0d12a0eecb3f3218b616a154c50ec2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f37aca09d0d12a0eecb3f3218b616a154c50ec2b"}]},"branch":"refs/heads/master"},"c26a697e55c92b3f445dc1aebe5ad842d939af53":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-19 17:27:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/2"}}},"commit":{"parents":[{"commit":"68bba0910d2925bfcf59e93bf5301c72638265a6","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68bba0910d2925bfcf59e93bf5301c72638265a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-19 17:27:38.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c26a697e55c92b3f445dc1aebe5ad842d939af53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c26a697e55c92b3f445dc1aebe5ad842d939af53"}]},"branch":"refs/heads/master"},"dcf1b74ac252b0d8faa92a5d3c520320630fbb21":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/3"}}},"commit":{"parents":[{"commit":"f470a03134260472c135909052e094ba64118c7c","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f470a03134260472c135909052e094ba64118c7c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:51:11.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcf1b74ac252b0d8faa92a5d3c520320630fbb21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcf1b74ac252b0d8faa92a5d3c520320630fbb21"}]},"branch":"refs/heads/master"},"67a7a9123f0b70815e1aded8a6d61148691e431d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-25 13:02:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/4"}}},"commit":{"parents":[{"commit":"cafa49a67449ca797091e6c362c0d696f02157b4","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cafa49a67449ca797091e6c362c0d696f02157b4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-25 12:43:38.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67a7a9123f0b70815e1aded8a6d61148691e431d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67a7a9123f0b70815e1aded8a6d61148691e431d"}]},"branch":"refs/heads/master"},"12cf46633118a82ebe6cf574c524157140aa4a27":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/5"}}},"commit":{"parents":[{"commit":"952273d208d6bf2fb8dc6f2b49dd9c7e968de268","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/952273d208d6bf2fb8dc6f2b49dd9c7e968de268"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12cf46633118a82ebe6cf574c524157140aa4a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12cf46633118a82ebe6cf574c524157140aa4a27"}]},"branch":"refs/heads/master"},"baaaaee237670e37ac671f26bb8dd73e83b92237":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/6"}}},"commit":{"parents":[{"commit":"5d5bf44f526e3ea0a67981329833638d5969d6d5","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d5bf44f526e3ea0a67981329833638d5969d6d5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:00:19.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/baaaaee237670e37ac671f26bb8dd73e83b92237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/baaaaee237670e37ac671f26bb8dd73e83b92237"}]},"branch":"refs/heads/master"},"c47456889feaf49dec04f63fca56aa4bd18a0387":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/7"}}},"commit":{"parents":[{"commit":"92e561c775433d598f9bb0e303ed318f0af41f5e","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92e561c775433d598f9bb0e303ed318f0af41f5e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c47456889feaf49dec04f63fca56aa4bd18a0387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c47456889feaf49dec04f63fca56aa4bd18a0387"}]},"branch":"refs/heads/master"},"59fc886d9c555770a9ce94d0eb99651372d74436":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/8"}}},"commit":{"parents":[{"commit":"5cafe2899f923ca36e3fffa0b25bd6deca2c5702","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5cafe2899f923ca36e3fffa0b25bd6deca2c5702"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 13:02:42.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59fc886d9c555770a9ce94d0eb99651372d74436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59fc886d9c555770a9ce94d0eb99651372d74436"}]},"branch":"refs/heads/master"},"1f917e70678fe2908a2c1c7b82173d28c88bc7dc":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/9"}}},"commit":{"parents":[{"commit":"a995a2b21e638235574c002d5fa90be2e2e8d2b0","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a995a2b21e638235574c002d5fa90be2e2e8d2b0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:47:28.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f917e70678fe2908a2c1c7b82173d28c88bc7dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f917e70678fe2908a2c1c7b82173d28c88bc7dc"}]},"branch":"refs/heads/master"},"a693e055a458d4628c8bdfffbf51d7e791f29c70":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/916468/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/916468/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/916468/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/916468/10"}}},"commit":{"parents":[{"commit":"66008dd93ce0105089edd537b4b1d82f66b3d688","subject":"api: Keep track of extension controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66008dd93ce0105089edd537b4b1d82f66b3d688"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:29:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:53:11.000000000","tz":60},"subject":"api: Remove AdminController controller","message":"api: Remove AdminController controller\n\nWhile theoretically increasing duplication, this makes the controllers\nmuch easier to read, actually *reduces* the lines of code, and makes our\nlives much easier as we attempt to apply schemas to these things.\n\n  $ git diff --shortstat HEAD~\n   1 file changed, 126 insertions(+), 149 deletions(-)\n\nChange-Id: I0c8b299d01262a7b039e3d74f8f7464d80c87bb6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a693e055a458d4628c8bdfffbf51d7e791f29c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a693e055a458d4628c8bdfffbf51d7e791f29c70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
