)]}'
{"id":"openstack%2Fcinder~916731","triplet_id":"openstack%2Fcinder~master~Ib2a4db3250ed2036ca68adc1bb87ef2c048f0bc9","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-06-05 10:50:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib2a4db3250ed2036ca68adc1bb87ef2c048f0bc9","subject":"DNM - Do not merge - HPE 3par","status":"ABANDONED","created":"2024-05-06 05:24:35.000000000","updated":"2024-06-05 10:50:41.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b47590d0f1fa59d50ddba73b9d5b01ee618f410b","_number":916731,"virtual_id_number":916731,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-06 07:42:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-06 05:34:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-06 05:49:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-06 06:58:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-06 07:01:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-06 07:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 08:48:45.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-06 19:07:25.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-07 03:21:41.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-05-12 09:44:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-17 00:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"bc6b304cb4cdc1363e6f0dc17cfe608581584a23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-06 05:24:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5fed2b229bcd3ffa02b0337324f668fde725934","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-06 05:27:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb6db1cc35da713a6ad8e54aeb2b70acf4fab36a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 05:34:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e3afcf960c4366895e76f56a10df24\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c572d1a4d12e4c0b923f090c9c94b6db : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe07294342ba4de084876a45ba666087 : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d8e0dcf35ce3803f993a65423a47b2491d41327","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-06 05:49:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916731 : FAILURE in 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bc779b7e1b8c19d6d07b5a8cd1665f2d02935780","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-06 06:31:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916731 : FAILURE in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5abfae78d318c06a4928da473b353babd37c0583","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-06 06:58:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/916731/1/check/DellEMC_Unity_FC/ad9f595/DellEMC_Unity_FC/1808 : FAILURE in 1h 30m 07s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/916731/1/check/DellEMC_Unity_iSCSI/10ec1ef/DellEMC_Unity_iSCSI/1836 : SUCCESS in 1h 33m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e092fda49fc22978f950737886a24660015971d4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-06 07:01:30.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9ee685dbdac24203bc2095021e416888\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/814a95b13b5244fdb9edc95008c8d84f : FAILURE in 1h 28m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e30d7dda321c0bd31f2c665094d1856b237a25f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 07:42:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09d7fa3be8104b7bbe23df22ff16b0a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f5d292bee04cbf9f24c7d39ebbadad : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61a6ebcd2c384f559579d9efec933377 : TIMED_OUT in 40m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7526d9fc73f44d418f8aa6d7fc2a93bd : SUCCESS in 10m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0eaa2434bc23466bb9bf5b6da432e861 : FAILURE in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/480fa345d14543f1a6f107a9b97c0fa0 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/3714a133fa2a404f8d4ceba1ac47463d : SUCCESS in 44m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7bbfe9cb5864ae898edc4ea93873ac4 : FAILURE in 1h 02m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d80e62f5f5d4495cb83519d7a03a15e1 : SUCCESS in 1h 10m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ef23b551ee16472588653cf5736b8500 : SUCCESS in 12m 07s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0254ab030fe047a6ac5ae8a4a5b77ec1 : SUCCESS in 12m 44s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f42b6afbc1294ffc9a668ccb070bfb66 : SUCCESS in 4m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a81a03755c7e4f66bf8b97399d49d846 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cafcc59c268c4843bbe5a36aacc03908 : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4dce1c0608a9490a83cbc79da575f0b6 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be8cc307e18d47ff8d95d9298d34f1c3 : SUCCESS in 28m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/928e5786c9e0491fa8509b793cfca11f : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/735ec3bcda8d46e0aee408f2a7db270a : SUCCESS in 1h 31m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf6c8ea9a5fb4d2f9c5fbbef627a4ecb : FAILURE in 2h 06m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/528a91ebf7be4fb4962aec82c172fefe : FAILURE in 1h 36m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/68b4612148e940ebbf8b07f6341a74a4 : SUCCESS in 1h 31m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/18a55dd760a245e6977aac38d881248b : SUCCESS in 40m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d024cc6a92ab410eb91e1ed2a0451a86 : SUCCESS in 49m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ca20316bfda4b6f9593138fed416262 : SUCCESS in 1h 28m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ff7f31396e6047619c7bf0b24e07d73c : SUCCESS in 49m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b03b5c1384cd428cbf0c388492c3ccf2 : SUCCESS in 1h 21m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/aac6977d29fd485b87608d25ae6cfb45 : SUCCESS in 1h 05m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f86a62a11007495586d26ad228f08257 : SUCCESS in 1h 20m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f852344890bc42bfa7ac72bccf32ea5d : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1620c25d31ff0e5f2e959c6dceab9d18f36875a3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-06 08:48:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/332/.","accounts_in_message":[],"_revision_number":1},{"id":"260974a8d5016196eedc2e8502aff642bd1e132c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-06 09:51:15.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/22/.","accounts_in_message":[],"_revision_number":1},{"id":"54ef3453c3bc2a8ba30a9bcab527eeb629bcc0c6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-06 12:01:59.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/23/.","accounts_in_message":[],"_revision_number":1},{"id":"a05c41c26d2fd4fa9199cc5ef314861c46adfffc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-06 14:04:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/24/.","accounts_in_message":[],"_revision_number":1},{"id":"419f29f9b8c5e5bd959d34de047eabf81392ea37","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-06 19:07:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/916731/1/20240506232039/NEC-ISCSI : SUCCESS in 4h 16m 46s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/916731/1/20240506232039/NEC-FC : SUCCESS in 4h 46m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8657fb4e16afaf9b78f41804cc84241131963ffb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-07 03:21:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/916731/1/20240507090905/NEC-V-FC : SUCCESS in 3h 12m 11s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/916731/1/20240507090905/NEC-V-ISCSI : SUCCESS in 3h 12m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0bc85ac71a75129ff8a86a7a540682209087adf8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-12 09:44:00.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/916731/1/check/DellEMC_PowerFlex_3.6/0d55cde/DellEMC_PowerFlex_3.6/1297 : SUCCESS in 1h 40m 28s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/916731/1/check/DellEMC_PowerFlex_4.5/3d64741/DellEMC_PowerFlex_4.5/224 : SUCCESS in 2h 22m 12s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/916731/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/7d76730/DellEMC_PowerFlex_4.x_NVMeTCP/170 : FAILURE in 2h 54m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"5df3790871a0ce50871744067e7676535dc811d4","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-16 06:22:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"534478b4d45223c5b4da2c77d78a17f6ead05800","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-17 00:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/916731/1/fc : SUCCESS in 0h 47m 28s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/916731/1/iscsi : SUCCESS in 0h 53m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b47590d0f1fa59d50ddba73b9d5b01ee618f410b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-05 10:50:41.000000000","message":"Abandoned\n\nthis was test patch. Now it is not required.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c86c8874fb5b5cc4d8529af5513fd6f899c328d","revisions":{"3c86c8874fb5b5cc4d8529af5513fd6f899c328d":{"kind":"REWORK","_number":1,"created":"2024-05-06 05:24:35.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/31/916731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/916731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/916731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/916731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/916731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/916731/1"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-06 05:27:03.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-06 05:27:59.000000000","tz":0},"subject":"DNM - Do not merge - HPE 3par","message":"DNM - Do not merge - HPE 3par\n\nThis patch is created for testing only.\n\nChange-Id: Ib2a4db3250ed2036ca68adc1bb87ef2c048f0bc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c86c8874fb5b5cc4d8529af5513fd6f899c328d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c86c8874fb5b5cc4d8529af5513fd6f899c328d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
