)]}'
{"id":"openstack%2Fcinder~916788","triplet_id":"openstack%2Fcinder~master~I0164582af81daf49b7391944940d2f8c187a9d42","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2025-04-11 20:44:27.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-04-11 20:44:27.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-04-11 20:44:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0164582af81daf49b7391944940d2f8c187a9d42","subject":"trivial: Enable E275 harder","status":"MERGED","created":"2024-04-23 12:51:48.000000000","updated":"2025-04-12 02:56:01.000000000","submitted":"2025-04-11 20:44:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916788-openapi","meta_rev_id":"ae31150066ec6804596073936e7df9330c443e2f","_number":916788,"virtual_id_number":916788,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-11 20:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-02-21 15:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-10-24 16:39:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-10 15:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-10 15:29:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-23 13:22:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-23 14:21:24.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-23 15:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-23 16:06:27.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-04-23 18:20:21.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 15:25:04.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 23:18:40.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-25 09:29:39.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 12:19:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-25 20:58:28.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-05-09 20:55:03.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 11:21:38.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-05-18 02:55:44.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2024-05-18 03:17:42.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-18 06:06:07.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-18 12:18:37.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-19 00:30:37.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-18 14:14:57.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 15:59:25.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-18 16:30:08.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-18 19:28:24.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-15 21:00:09.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-11-16 00:44:04.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-01-13 17:18:07.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-02-21 15:15:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-04-10 15:29:44.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-04-11 12:31:26.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-04-11 23:21:37.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-04-11 23:38:38.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-04-11 23:54:43.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"9a7a97080d4a647c76f30fc2e2aefb406fb289f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c5a53d941fbe6575ee7c6b3b358d4345bc8bad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:22:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/643d3b43aa0b47b0a53230deaa07ce09\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce0c98bdda9545f99cff514d72516f9d : SUCCESS in 26m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/91cef28b69e842d2884f20568e03af9f : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9464f4ef7369a8f808cdbfc0734ead86463eea97","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 14:21:24.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"fe551d9f626e447d5367d3c31212ed9d5834a7bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:14:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c16187c01c8430db9bb8736642974b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe081a66a9194f2fa4787626cdc68ba5 : SUCCESS in 8m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab6d18b8b8ae4360b0a019439437ba01 : SUCCESS in 11m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25810397bc6741f68111ddb66d6e9803 : SUCCESS in 12m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b7d9cde251c4d398d17bdc2f9f386b4 : FAILURE in 11m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b18327e0659459c975b6a0512d2a516 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/678e37feee8a4f01ad92a0dcef741fd4 : SUCCESS in 1h 11m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a261865469545c5b13a66a81a18bd97 : FAILURE in 1h 19m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7156a4afe8349a280ce0d3a623f869e : SUCCESS in 58m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a772e33c75f54a82b969eea18d6972e2 : SUCCESS in 18m 57s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c3658cf0a2674902b50d04854a607d41 : SUCCESS in 12m 17s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/75c7631b190948ea8d9816def461811b : SUCCESS in 6m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d1a695f6d4743dc9b7c0b07fc6dc209 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aa1bfd28f8c94f3aa23ad4d1adf0339f : SUCCESS in 6m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/022d134b1e3c42069b27d990f0b50c08 : SUCCESS in 7m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/664f704061e54d498b3d517c6c7ea69a : SUCCESS in 45m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4022898c49e341bf97bde9dad58f3af6 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dd036abbd8e34e5bbf7b3490dce1d9d4 : SUCCESS in 2h 12m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a102b55e08e34a97bdd7583cd977291a : SUCCESS in 2h 08m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0579ca62a03f418daff6681502e523b2 : FAILURE in 1h 26m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2480a3f132cb4a1a8bf7ada4b36ab13f : SUCCESS in 2h 10m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a3af8a2b5e7e4e0992e6fb7a9b24641d : SUCCESS in 29m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c5c8d52f35f4ff89fa8e1c984b18508 : SUCCESS in 1h 19m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/075d600fa0b84628b04855f0a9b64778 : SUCCESS in 1h 27m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/07327b494a8e4fa69f877d4646884c74 : SUCCESS in 32m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3a7178a93bee4dd7965e6064870376bb : SUCCESS in 1h 26m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3accb8905c2e47e7b8560b6007e2370a : SUCCESS in 1h 26m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/53de37d878d64a53bcbea1754b7b3c81 : SUCCESS in 1h 14m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b6004d7fad34cd4b48eb562f7bba607 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"626a0db5b8a3f73fe74c9cc6443f8fbe53e97b7a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-23 16:06:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cefe914e511faf24b3aee84eca337127d5d9ae19","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 18:20:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916788 : FAILURE in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0afe2c48dea97b2a5eac8f4894632ab66c7a2388","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-24 13:13:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83c4e0944c351aac44b6387e4d4a8483e469ce6c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-24 13:13:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3aaa40c47fde48ff2b54f70c77d5c4a886efe5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:32:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d02b0ecae1084ca48a0d448b90df27ff\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfce04b64b054eed9493b79465490869 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b4a029be4df94e1492ec0cca5c34a1bc : SUCCESS in 17m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67f288d55c44f905918cca2a4d57fca4b8f42248","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-24 14:56:08.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f2ad0ff87cae483191c6de29ece6fff8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/08b9555e58084290bca2c3d44618dc10 : FAILURE in 1h 29m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d0e4f4c107bed59ec01f6e3839072449d4257c83","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-24 15:25:04.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/88/916788/1/check/fujitsu-eternus-dx-iscsi/6e4efd2 : SUCCESS in 3h 20m 36s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/88/916788/1/check/fujitsu-eternus-dx-fc/9389f63 : SUCCESS in 3h 08m 33s","accounts_in_message":[],"_revision_number":1},{"id":"66bc66b0025afe4b1cd4078192a2123cb46bc1ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:49:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bd1fb65dc06400a83eae58df54406d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf6340db1f1483883d9a4cd885a6d6b : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7035981cfc874ad0829234de87e6e412 : FAILURE in 12m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0c3ced8c0d040c1b8bdcb465482b66b : SUCCESS in 13m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5418a1f06560456d82a4ed55551404a6 : FAILURE in 10m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a63bfba4c08246a2aec0e50701b276a8 : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/ce0b01d6d76746deb321f12cc7391f5f : SUCCESS in 1h 02m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f91ddae2bfa64a0895c5b16d5a24978b : SUCCESS in 1h 33m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92c29df4aa8d48ca8307ca88cdf09403 : SUCCESS in 1h 01m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/151796e023244f4c8aeedfe19bc62a4e : SUCCESS in 15m 08s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cb83b2bba54443f4b3a2ffde09e1bb00 : SUCCESS in 13m 18s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/977f77979b254bb58d4471fcbbb023b8 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/855faf9b76ae4c3a8be9457e6a3e1e56 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/667b5b26c155433eb7b2c0e06244f216 : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/093b71cb23784932ab7e8d45af2fb6f5 : SUCCESS in 7m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1e97fb3c27bd4366b94277784156cc2d : SUCCESS in 1h 01m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59cd78bdc55c488a973e607ec5434684 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cca83bc38bab42e6a57d88805eb60d58 : SUCCESS in 1h 38m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/82f0051c3a884df5ab35dab2d4852148 : FAILURE in 2h 13m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/358dd2b122104ab0aa5a09b4c04586ea : SUCCESS in 1h 30m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/379236e5466b4c7c8feb466abf0ebe27 : FAILURE in 1h 42m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eab79dcd6082453c845c398393da3e03 : SUCCESS in 29m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/47c4eea6e32d4cecb07bf997e260bf5b : SUCCESS in 1h 09m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/db51fbe1abff42629815e40d264f5e48 : SUCCESS in 1h 29m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/04594dfe4aaa4cbc9c9abf9d7ba05950 : SUCCESS in 31m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8df055bad02d4cb492cb7db40c0e4c39 : SUCCESS in 1h 09m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d2ecde4dc9364ad687929c352b2a69ae : TIMED_OUT in 3h 31m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e36e81f0a4d84aef88ddf9fe8d849357 : FAILURE in 1h 49m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0a163302cb14d22bffdb4e61c6198db : SUCCESS in 1h 25m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e37b29eae82a8b66c020903d60fd9abc9be54b69","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-24 23:18:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/916788/1/check/DellEMC_Unity_FC/28d904c/DellEMC_Unity_FC/1757 : FAILURE in 1h 29m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/916788/1/check/DellEMC_Unity_iSCSI/0fff529/DellEMC_Unity_iSCSI/1787 : FAILURE in 1h 56m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"57a3765c215c832b798f6b62e7fc0dbbab263675","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-04-25 09:29:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/88/916788/1/ : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"33831608247fa41036dd029c5db03b9778f0dbcc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-25 12:19:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/297/.","accounts_in_message":[],"_revision_number":1},{"id":"2a00aa13bf55e4b1e3fa40b83520f4c1c15e34ba","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-25 20:58:28.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-916788-1/FC/ : FAILURE in 40h 23m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-916788-1/iSCSI/ : FAILURE in 32h 41m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"b007c7c87509b31aa1bfe3c68724a5df54765881","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-26 16:03:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19fe6dd26649edb23afe471dbcdd5a8218c9a33d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:35:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a1993a9c0544bba44b4b24b6a61408\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8fb469faff8f462594709384c64c086d : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb7d62f23c42472f9ebf91a92deaf46e : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df7178a5aa0503959a50cd4842891496ee59f61d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 18:52:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18654d2af62e4031ad31faaa8a52dddb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036d46c48e6e4f8f9b957790d850c077 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56c33fc4bec403499f7dfce745547fd : SUCCESS in 11m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88665dad1a2b4f26963c9dcdeda8c4fb : SUCCESS in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49ad8f6f7c6e4eac91adcacf97fe3c32 : FAILURE in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d03064ab09f4f99a67ee0829f87704c : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/1c2fbe200b1d40a2bc1746a75f37b141 : SUCCESS in 1h 04m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e079e2378c646c68d93073d90b173a7 : SUCCESS in 1h 33m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4ba3303b7da4332a2cc640151223816 : SUCCESS in 54m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/492e42b3860540b3bcd7b036851d0645 : SUCCESS in 16m 42s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a036e3a8244a4f4489ab8516f4f25e47 : SUCCESS in 12m 34s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c1cfb9734de74cb3aad9907b665d719b : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/07c72040085349708c94f27b84be2be5 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e85ed6844b0f457e9bd457318a57dfec : SUCCESS in 5m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2cfe33a551974c0280e4004a397afd73 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/755c506a39914dbe88ce3ffb2d68f542 : SUCCESS in 37m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/05201700a4174b15863623f8b983b247 : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4b4cd18b3a0940818762345bffcbf17f : FAILURE in 2h 31m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5694520e7af84ad49f6cd705f66fc52c : SUCCESS in 2h 14m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/263869863eb4444ca895a5fe5716d337 : FAILURE in 2h 14m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b6e048f9c24346998b11b361c0df3305 : SUCCESS in 1h 42m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/284cbee0e67045ce834eb53c0a165c14 : SUCCESS in 45m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1a6d52fcf45340a3852516c769d374f9 : SUCCESS in 59m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48c769129062461587ed717fd37d788c : SUCCESS in 1h 25m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69cd3168c77c4c2c8632ae0fe203c0a7 : SUCCESS in 52m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/43185cb0d830460f99c66baec3455f65 : SUCCESS in 1h 21m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dc3dc7b6696e4f008cb837d0d07353fd : SUCCESS in 1h 30m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70a5657a0bdc4bae97d24d9899f75011 : SUCCESS in 1h 35m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0faf7d9acd67452c8e96c65cee01f7a6 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6b3a453920708501e421d0ddf1932fa6d1e7ed9e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-27 01:32:35.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/667ae23cbefb45c3823e25dfea2ce468\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/702165fcaac64a00b3627e4258df8d70 : FAILURE in 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"230bed595c079b436c7f63a303b84045c3c96432","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 20:55:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/88/916788/1/check/DellEMC_PowerFlex_3.6/934855a/DellEMC_PowerFlex_3.6/1259 : SUCCESS in 1h 46m 05s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/88/916788/1/check/DellEMC_PowerFlex_4.5/f6bc359/DellEMC_PowerFlex_4.5/195 : SUCCESS in 2h 00m 55s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/88/916788/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/86a4b3d/DellEMC_PowerFlex_4.x_NVMeTCP/135 : FAILURE in 2h 54m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"d3619d56116f70419a02d79bc13cc37fe8f0aec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee9f25a01bafc53bbd10b062f87473e97c90aaee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:13:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09c35bc9523749ea9951576a4adcdb1e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c79e65062f2459da55e81db24ae6be2 : SUCCESS in 14m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/642b02e1517e41368a1823ba80111197 : SUCCESS in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"689f6455544efac3d17f3a45bf6b58c6baead593","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-13 15:19:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d74ea2718eb169cb7d62ff7f97f060a71074cac9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-13 16:40:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916788 : FAILURE in 47m 53s","accounts_in_message":[],"_revision_number":2},{"id":"51d0ed80802b700c9cbbe0df3d9815f30d2cdd8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:59:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93e5051284e44d6cba9ceea8f6def7ba\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/805ae089245741268e473fb96c924099 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70815f3b823c4cedb8ba051fbb716fd7 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eefb0023c9a742dc9f202096e8b186b8 : SUCCESS in 9m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f19f75ebf424761a068dc7dd7d173c5 : SUCCESS in 12m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07d8d4b22aa74da1bb298695c25000f4 : FAILURE in 11m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360b7b658d104c4990c312e321b698e7 : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8753ad9506d546e09b66878f486556da : SUCCESS in 1h 24m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/511da4eb1af94c8380d25205625757d7 : SUCCESS in 1h 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/749eec9b34b841d3aff2e73007f8ef44 : SUCCESS in 1h 07m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a451c8aa7f314abfa7af0ec82776fa6c : SUCCESS in 13m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f061a0d76ac4110ba29e2a2638b9672 : SUCCESS in 18m 05s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dbbfff4ca42545669483f31dac73978d : SUCCESS in 14m 49s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/31bba8f1b8b04e3890a1aaac88dbe626 : SUCCESS in 6m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d2015666c42487794a9a556f390c550 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07a087beb77848acb5a5bf3f48914572 : SUCCESS in 6m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/89511eecd5664014832ddd713a8a1747 : SUCCESS in 6m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c071d447c94d4aa9ba9ea7563b15d156 : SUCCESS in 42m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f160c6c6177a4188ab9c2a66d23081bf : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/843ee558ad004d5ca58f4ad2adbeb115 : SUCCESS in 1h 48m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a176f0a9591b4900a141ea06527988d5 : FAILURE in 2h 01m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/359a8572b6e243aaa1e878f157967e08 : FAILURE in 1h 35m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e93b04ebbb6d40b7a2179fe58f6cf67e : SUCCESS in 1h 45m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/15da36648e1143d8bf8f9711eddc314d : SUCCESS in 39m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/72cf4da37c874d2fbedf10ea3886d509 : SUCCESS in 1h 09m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3acd3a204f1c4e0fbef7a59554f87683 : SUCCESS in 1h 23m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d454b8f916ea44a6a496bc552e76191b : SUCCESS in 47m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0478c18363a0446aaa18cc99ebbeb64e : SUCCESS in 1h 23m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/13130bedb70c4183b4891de5fe3ae7bb : SUCCESS in 1h 30m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/194446c5c4ee4107b16103ec7f6fbbfd : SUCCESS in 1h 35m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a59b69145cd42ea8e7c6b4b676b7ed4 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b19e02f842a69517d160c7cfefa9e2aa2d6696d6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-13 17:34:05.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/328b1a12449c4350866c737fbf0aa17e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b80eb3752a354ad697136beb73baac84 : FAILURE in 1h 27m 14s","accounts_in_message":[],"_revision_number":2},{"id":"679a09ec7a206f70edd5c2fb94f82854e025de4d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-13 18:41:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/363/.","accounts_in_message":[],"_revision_number":2},{"id":"7d50a201e3fc3451c0a367ac6284a4264fbb5c3e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-13 18:55:33.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/916788/2/check/DellEMC_Unity_FC/3aaaca2/DellEMC_Unity_FC/1842 : FAILURE in 1h 52m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/916788/2/check/DellEMC_Unity_iSCSI/cd73135/DellEMC_Unity_iSCSI/1870 : FAILURE in 2h 14m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"61f9bc8c6f346a402c66a38f349569017be86691","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-13 20:16:26.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-916788-2/FC/ : FAILURE in 4h 53m 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-916788-2/iSCSI/ : FAILURE in 4h 54m 51s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"bd89fe541e471892a0c22fdc18b3919d71dfa63b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-14 18:41:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/88/916788/2/check/DellEMC_PowerFlex_3.6/4717634/DellEMC_PowerFlex_3.6/1324 : SUCCESS in 1h 40m 48s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/88/916788/2/check/DellEMC_PowerFlex_4.5/19aeda4/DellEMC_PowerFlex_4.5/249 : SUCCESS in 2h 12m 40s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/88/916788/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/c329c4e/DellEMC_PowerFlex_4.x_NVMeTCP/196 : FAILURE in 2h 55m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"f8a330b63f6f5b048ff3de1f708ea52e2a7e4b1c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-15 11:21:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/88/916788/2/20240515092158/NEC-V-ISCSI : SUCCESS in 3h 22m 03s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/88/916788/2/20240515092158/NEC-V-FC : SUCCESS in 10h 59m 36s","accounts_in_message":[],"_revision_number":2},{"id":"83ba28a882492acef02547911abba4847d1b9ae7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-05-18 02:53:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"125e758356d18432d9320dd3410f2e03daccf387","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-05-18 02:55:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b77e0a681e100eda14e4b57df36ab405b5f2f1bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 03:04:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36f01995f1db4ce38ae835258a316841\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3103c37dd2234c84826ecb5a97b08536 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/14801324d89b43e0aceedea694ccedd8 : SUCCESS in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fe4c3dbe817d9ebd11d22d954990a3e7210ddae8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-18 03:17:16.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a2a798a0cba74959bfda15cdcead0edf\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ae11eb6dd1a24d3180973ac1d49ae748 : POST_FAILURE in 21m 03s","accounts_in_message":[],"_revision_number":3},{"id":"079d6e1a41e1a8e9e614157882d07f157af75e7e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-18 03:17:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916788 : FAILURE in 22m 46s","accounts_in_message":[],"_revision_number":3},{"id":"acf69ff65f59cfa948c96004f3d3280ac6c7d034","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-18 03:40:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916788 : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":3},{"id":"2eded3bc7714071bbfaaac69cc6c0eb44c5a8e75","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-18 04:15:26.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/412/.","accounts_in_message":[],"_revision_number":3},{"id":"1353cb842d655d2299722b7e8854d07cdfa58770","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-18 04:57:39.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_Unity_FC/0c81463/DellEMC_Unity_FC/1884 : FAILURE in 1h 48m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_Unity_iSCSI/95e27fd/DellEMC_Unity_iSCSI/1912 : FAILURE in 2h 03m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"be542ddd2f68154d6c7d6931158e4b4d696bc2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 05:24:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d390783a05db4e4eadcc5c8318c8b8c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/359d48e275424ba887a460b79646b84d : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c57850a3cde4424b977838ee3ba3419 : SUCCESS in 11m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b84c48dbeda451c9782046b7d618d3e : SUCCESS in 11m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32573a6c5e364b68831676895537ef87 : FAILURE in 12m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98991f9c10c9411ab124d009d984b8cb : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/0f86cc7d184247dda20a02bd1bb5b812 : SUCCESS in 58m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1f79940e9aa422cad05a9677b711f5d : SUCCESS in 1h 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70325d2858e54828b3e16e903546a990 : SUCCESS in 1h 13m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5416f685c352442086e1bfc7810683d7 : SUCCESS in 14m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/15daa518f6664cfcbdb77696c1c425ed : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8edd5e04acf54a55aad5663841d85544 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0fdbbc48cfa643da88a1f269698c5626 : SUCCESS in 8m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4052af296a7f4d77bc98b9647eba0b43 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8495c609c7f5466cad5bf92938919010 : FAILURE in 46m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f9cfe16146654f6ab406c50cef9f501a : FAILURE in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f97f32002444cd088e9f7dfb5e9b149 : SUCCESS in 1h 46m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7af0c1bfe1284bc98eb8b48fec277951 : SUCCESS in 2h 24m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02612dd85fea495ab52ca3dc337bbba6 : SUCCESS in 1h 14m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/78d298a070784c42aebf97cad7a94493 : SUCCESS in 1h 39m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5bd74d51cc644d41a5b14590c3244832 : SUCCESS in 36m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f83980d3ce3b45a9976e873905c92934 : SUCCESS in 1h 03m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1aa7c4b9b81d4ec286258fd7f3aef0de : SUCCESS in 1h 15m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a1ed90119ea840458fe5d54a237eaacc : SUCCESS in 40m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/37d3a2be6335488ca19dad094a0eceb9 : SUCCESS in 1h 18m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2a5e4b9b34724c879694210dc384dae1 : SUCCESS in 1h 29m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/551392ac747349cc99eb08849b9af77f : SUCCESS in 1h 33m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4987fdc93c24cdeb9ce47d10f2a2f45 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":3},{"id":"390fcd8544a41415fd5ed2039f877cb8283685b5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-18 06:06:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/916788/3/upstream-check/cinder-tempest-plugin-ontap-fcp/e7d94ed/ : RETRY_LIMIT in 3m 09s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/916788/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/b0f7dc7/ : RETRY_LIMIT in 3m 36s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/916788/3/upstream-check/cinder-tempest-plugin-ontap-nfs/f0dda77/ : RETRY_LIMIT in 3m 47s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/916788/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8af3864/ : RETRY_LIMIT in 3m 38s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"3a8152fec08963b25ecaeeb94d26457f218d18ab","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-18 08:43:22.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_PowerFlex_3.6/a3e4d4c/DellEMC_PowerFlex_3.6/1383 : SUCCESS in 1h 39m 36s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_PowerFlex_4.5/ae550b3/DellEMC_PowerFlex_4.5/303 : SUCCESS in 2h 13m 40s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/e60c323/DellEMC_PowerFlex_4.x_NVMeTCP/248 : FAILURE in 2h 55m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"fed693523b64679c2ebeb7e8c832ebfee4883144","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-18 11:27:32.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-916788-3/FC/ : FAILURE in 8h 33m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-916788-3/iSCSI/ : SUCCESS in 2h 8m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"7d684c580f3f4f5a795b64021882abc29230f5bb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-18 11:48:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_Unity_FC/c28e87f/DellEMC_Unity_FC/1887 : FAILURE in 1h 48m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_Unity_iSCSI/61758b7/DellEMC_Unity_iSCSI/1915 : FAILURE in 2h 27m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4f8a87e6798239138a5509303838e47b639e9ac2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-18 12:18:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_PowerStore_iSCSI/45bf05e/DellEMC_PowerStore_iSCSI/1329 : FAILURE in 2h 54m 19s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_PowerStore_FC/a5721eb/DellEMC_PowerStore_FC/1512 : SUCCESS in 57m 39s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_PowerStore_NFS/e32c71b/DellEMC_PowerStore_NFS/1023 : FAILURE in 1h 11m 53s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/88/916788/3/check/DellEMC_PowerStore_NVMeTCP/a0de9f9/DellEMC_PowerStore_NVMeTCP/877 : SUCCESS in 1h 40m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"00e9b1849ead25e6d7fc6c733dbb90ba31c2a873","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 09:56:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b4e9b4554e634402fc01b9c4933f2d9a5de40376","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:01:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"62e5efbb558f534c57bd331087b7280bc1acd1a7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-20 15:11:50.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/88/916788/3/20240520162731/NEC-V-ISCSI : SUCCESS in 3h 35m 31s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/88/916788/3/20240520162731/NEC-V-FC : FAILURE in 7h 44m 16s","accounts_in_message":[],"_revision_number":3},{"id":"86ac803fb8ed1a7e7199b7661bb46f63f15e5905","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-19 00:30:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/88/916788/3/20240719034350/NEC-ISCSI : SUCCESS in 5h 17m 24s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/88/916788/3/20240719034350/NEC-FC : SUCCESS in 5h 46m 45s","accounts_in_message":[],"_revision_number":3},{"id":"5989e3161cb60319012790aebc1cdde37b177633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"26c981da1a340aa2aec4f4de75f86276b1993288","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:14:57.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/579/","accounts_in_message":[],"_revision_number":4},{"id":"93bd0ec0afd14b9984b89115889283a77e4874ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:14:59.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916788/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916788/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"8179ee97a1e46b6d93a394522b72032307430a3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 15:06:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b3764402bc446b96e846d194f8c0d6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d673382c22943e1bae4f346b0d6dd58 : FAILURE in 26m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/560d0e5df897418ba611894837266cf0 : SUCCESS in 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"28d6d22bcdefc7e851360fcde7733476f1502b8f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 15:59:25.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a9f19e5178514c25b94e8351ce3596f8\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2dd97d5c63154a69b3dfa555744e6f8e : SUCCESS in 43m 16s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9460fd0fbf444f17b55d1cf499f36f8f : FAILURE in 49m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb9faf09c50b46889e1aeb94d0c8e980 : FAILURE in 1h 32m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18f3795602884dde9f89619d61b5b3ae : SUCCESS in 40m 51s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"6f0f989b12d048be4439cf306be465791934671c","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-18 16:30:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916788 : FAILURE in 42m 44s","accounts_in_message":[],"_revision_number":4},{"id":"e7e2b54549ddfd205b4c4bac232673e02bb6ada8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 17:08:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75dc1624d1284346be7da2f7b9e48535\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1862c330dc4a59b120603c58b91c89 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03eaef2518214b1cb69bfcb6327fc935 : SUCCESS in 10m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d93b428057f46f290a74e9403d8fcc0 : SUCCESS in 11m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ae1a9f9eb874e2a96807e0cf2269cc9 : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b650b301167e4194ba79c0bd74f8d159 : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/2fd64a5236bf423c8a22dcc6ebeff7d2 : SUCCESS in 1h 10m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c110c72ed8994771b9c528728a0b6ea3 : SUCCESS in 1h 21m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdf4dc48aa93406286eace86e08dfea5 : FAILURE in 18m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e83fc026f95b49f799e9f1a45b780937 : SUCCESS in 18m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/20282967ce22498ea3cbceeb058af34a : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32ea5288e06143aa873d85a6805f9740 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2f23a519ff04c00ad1eb4125fd89b3b : SUCCESS in 6m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/caf3fa32e9a941a695584f6965871b69 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e60a6017c31746d081d05819d8c064d8 : FAILURE in 24m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b75f821db3d43d08a5ebfd03938c153 : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/98c6ddccca494e7ebcc6737ebba84056 : SUCCESS in 1h 38m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/194fedce0d754c02a898adbbd335c463 : SUCCESS in 2h 41m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b1e44ead95ba44d0bc3daeb292a24bae : SUCCESS in 2h 23m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/308673ea569d481c8291b1461b5f34c1 : FAILURE in 1h 56m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7fa136f22b0d4f81a54569679a04bf21 : SUCCESS in 26m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e411156684c24f409b193e2ef60c7771 : SUCCESS in 1h 25m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/50bbf481c2a34c9fba7288bcffacc3be : SUCCESS in 1h 19m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/19e4b8c5219e45a2b4d1ba65355f8a38 : SUCCESS in 47m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c295be4f45774fdfa4dd667204544a46 : FAILURE in 1h 37m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ff53180e15af4e0699bbd3048ea973d5 : SUCCESS in 1h 40m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/234cb575d56b484dbae4f5157393a259 : SUCCESS in 1h 46m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c839f3dba6394341bdc6767dbd801709 : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":4},{"id":"164f037469d9b19fcc11080c82428cb225d2e8a2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-18 17:34:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916788 : FAILURE in 1h 00m 33s","accounts_in_message":[],"_revision_number":4},{"id":"a85d92d2d7fa8cac7c402c9cb3739632dcad553e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-18 17:44:53.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916788 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":4},{"id":"7af45af7cc273281663cdb28815812b3e8ac0ce6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-18 19:28:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/22aa1976adeb466fa1c2166b17bb4f94 : SUCCESS in 1h 19m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9ea5edfa45244103ba482f7eb6b32750 : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":4},{"id":"58d63c3a9a0c54a89b5743bc9063823aa9e9d763","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 04:57:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_Unity_FC/6186242/DellEMC_Unity_FC/2621 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_Unity_iSCSI/338e910/DellEMC_Unity_iSCSI/2630 : SUCCESS in 2h 25m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"41e621228b61a0986e408ab5c974c9a4cca21280","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-19 18:48:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_PowerFlex_3.6/19ef376/DellEMC_PowerFlex_3.6/2115 : SUCCESS in 2h 26m 12s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_PowerFlex_4.x/f0898ca/DellEMC_PowerFlex_4.x/999 : FAILURE in 2h 04m 28s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/7a68e5a/DellEMC_PowerFlex_4.x_NVMeTCP/939 : FAILURE in 5h 53m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"2cb7bbbe3dace8658a0fde09375941b779a9ec94","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-19 20:14:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1142/.","accounts_in_message":[],"_revision_number":4},{"id":"76cd3e249686dbf3f97607e7a036eb94ae918f7e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-25 11:45:44.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/916788/4/20240925090110/NEC-V-FC : SUCCESS in 3h 38m 52s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/916788/4/20240925090110/NEC-V-ISCSI : FAILURE in 11h 44m 31s","accounts_in_message":[],"_revision_number":4},{"id":"ea3503885ba1b12aac5d91a04b78c7fb68705d2e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-24 16:39:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"16f7a3a09d0b8a32e5f8ffbff2492f64271dfdd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 17:37:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12765e5eccca4b31832b5891957eab94\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e7e137fcde249978880ac22f491f2d9 : FAILURE in 21m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9524260c7b9741e4b07669d2d06929ca : FAILURE in 24m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e8cab3e8c6622b9ce05043a8ceaa5027e7a65b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:02:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6eda9048b8da4108901d7ba0b1b95346\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042b26800c674dfea56ecb5ffc598426 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e65a31d0b8945c2bed54a9de37670ed : FAILURE in 8m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d88f5aa2fb324955b48f9ee76e84dff0 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d881a15c17c4b408a58a04611b6f4e4 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/0b3b3daee4fa47b58217d977842434af : SUCCESS in 50m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ff7e70c3a3f474aaaec636a0455a5b0 : SUCCESS in 55m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/efea1b0f0f834bc6b10d9f839d523c48 : SUCCESS in 1h 38m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e13ce728f6a944be8408a082ff34e87d : SUCCESS in 38m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b7917be039664132894d7e653180a7da : SUCCESS in 17m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5126505129f41ff9a8531aacdab1513 : SUCCESS in 4m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d42cea26a1c243e888adfac9f81da395 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0eb298ca844e43b7b3f0b30b061cafde : SUCCESS in 6m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7d12b7506152428886de4096cb44d765 : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/113db2908c9c4f4eb53f3d3ceda10da0 : FAILURE in 24m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/583c5d2d143844f2bbc79a704fecc00b : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/372d0a3da6fd4879aff555cdd1fde2d2 : SUCCESS in 2h 15m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1eaeab6d6d584f62a9fbb5bc8d733237 : FAILURE in 1h 39m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d54a5e9a7c94a458fb3b9ccb0122854 : SUCCESS in 1h 30m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/40cd242caf344a93ab89c7e9ea646a2a : SUCCESS in 1h 43m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/849750ec1e554d6893782d073a4c28fe : SUCCESS in 28m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7971e8c2dd2a419cb0209423bdf1e723 : SUCCESS in 1h 06m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f661de56d91c449082f9a86d1490a98a : FAILURE in 2h 00m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/abd9195090f540b6a86793f6d8f037b3 : SUCCESS in 47m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/79c8c294d7954e46acd4461117a7bbb3 : SUCCESS in 1h 45m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/73c8932b9fb441d2a918af2ccfdd105c : POST_FAILURE in 2h 04m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab02bcd5bd994e6ea34c633201b9568d : SUCCESS in 1h 35m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5140f3c9f327412eaabef2915161f3c7 : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":4},{"id":"f6ce7afd2a25e10c74565e8b6550af621703faa6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-24 20:05:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/235fb62e256a41c3ac10deee7dcc6a31 : FAILURE in 1h 11m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1314f8cff69d449ebf6f6ba92d40be45 : FAILURE in 1h 15m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f3c85eeb42949bfffc76096800a94a02e7023584","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-11-15 10:40:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c4985e6aec05cce9c153490051fdd21b5ca8ae4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 11:12:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aae757a63954977841180441f0b9691\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/933fa0f1d60b47c9ac35f6cfb894cd92 : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42ec3731a83c409bb7fb157ffc8c9853 : SUCCESS in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c7bd727cc50282d7c8b9da7e8f544347e7482566","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-15 12:19:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3d6476a845534a058c6afc884a0f8356\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/20cd36216af54503a937a643badea660 : SUCCESS in 42m 47s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/22b62375fe2046cd984a86e6b249062d : FAILURE in 47m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3492bc8c615747ddbf45694c62929b8e : FAILURE in 1h 38m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/727cbaf75b62420fb3675ca40fc71d89 : SUCCESS in 37m 21s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ff589f1468d4c3a86f729670b2e508e : RETRY_LIMIT in 37s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5632c4efc1174032824fb85a336a802d : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c380876ee0554bea825331edaeab7cb9 : RETRY_LIMIT in 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f74b805f72884783b5be5e5394e4aa28 : RETRY_LIMIT in 30s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"3420af58f9d28c81d38f49c3d7ea57848ae1496b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-15 13:28:17.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0c032281f01640968defba9727d28399 : FAILURE in 1h 12m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8dc006c1c3ca47cd87f7942f0b2a3a5f : SUCCESS in 1h 27m 12s","accounts_in_message":[],"_revision_number":4},{"id":"b12a9a93578e4631fa3240c292cb611643179224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 13:42:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d21153f4000d45898f7d1898a51dfa0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c999af728934ccabc2aa745de115ef4 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85e61e608de74df48506966020d4ba67 : SUCCESS in 12m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78116e79c2f44d15be4a4fa1a55da52b : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f626ab42601a4961935d5549029fc63e : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/1c60277d0a2c42f4bc5c663344ffeb2a : SUCCESS in 1h 03m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fcbc79b40ef048beb4796023f7bd42f8 : SUCCESS in 1h 00m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/908222360ec848539c7ab7353dc255cd : SUCCESS in 1h 16m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d148d4ca4054228858a0c71d4f086f0 : SUCCESS in 1h 07m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9cd5dc79e8de418aaa94f07461e9dd94 : SUCCESS in 19m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3f70d72e55e0437cb6a8b12d54759100 : SUCCESS in 3m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef8ea8a03b0248b4ad2e72c63373d6b2 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ce1cc042c804a64b0516b26868d1547 : SUCCESS in 5m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ece431d7e0994eeeb6a31a5331435c14 : SUCCESS in 8m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1380d0529159493aadcb67fc3d4ee9e4 : FAILURE in 31m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97a992a493f9428fbd2df761cdcdd0dc : SUCCESS in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f41e4e66b2e4039a5d3b4a14cc83be5 : SUCCESS in 1h 49m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b4201b366a441359d0d92c10e96bab1 : FAILURE in 1h 10m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/099c646dcf414110aead78a93e1cdab3 : SUCCESS in 1h 57m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9fd11ba7be67472bb267aed4ba7363fd : SUCCESS in 2h 15m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0c581b54941449909b1a40789ed3be50 : SUCCESS in 34m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4c759d6ef0aa47faa8a7ea6a8326452b : SUCCESS in 1h 01m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/42d5e200ca64488b822d3af75f2265a8 : SUCCESS in 1h 31m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0fbecf8c543945eda395e6bab41b8ed1 : SUCCESS in 50m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4c38a83fb2e74342a2902411cb62e940 : SUCCESS in 1h 24m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2b43b34abd7e4fe2af5c56f957e70f83 : POST_FAILURE in 2h 26m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a691822583b43e6add50b2837625d98 : SUCCESS in 1h 35m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5627d316dd694c26928effbbe4c6ec82 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":4},{"id":"7e5184484eef12c2a31f1785b27c576c4a6464bd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-15 13:59:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/cec24d31c82849a9a52f28abd11dcb94 : RETRY_LIMIT in 4m 52s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"b3c3a86a840e7442399f547e523a20e3b84f8dc1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-15 21:00:09.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/88/916788/4/fc : SUCCESS in 1h 02m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/88/916788/4/iscsi : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":4},{"id":"2ce97b78690a11a65769788f65c3d9cee2e9b577","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-16 00:44:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/916788-4/Seagate-iSCSI/ : SUCCESS in 2:30:26\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/916788-4/Seagate-FC/ : FAILURE in 2:17:12","accounts_in_message":[],"_revision_number":4},{"id":"971e13f0f3fc264880aa88252a9c91d72a1f690b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-16 20:50:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_PowerFlex_3.6/2f3e3fd/DellEMC_PowerFlex_3.6/2437 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_PowerFlex_4.x/d23e0d0/DellEMC_PowerFlex_4.x/1285 : FAILURE in 2h 03m 14s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/e4765c1/DellEMC_PowerFlex_4.x_NVMeTCP/1178 : FAILURE in 4h 34m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"821efd4b0b3af508866b88d5e95c9ef014232136","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-17 07:00:44.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_Unity_FC/8599a08/DellEMC_Unity_FC/2889 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/916788/4/check/DellEMC_Unity_iSCSI/583415b/DellEMC_Unity_iSCSI/2856 : ABORTED in 6h 00m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b1b2825a3e20753565f2d695d3266aef7efb719a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 17:18:07.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2025_17_12_19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"67a555a8a4ed52926fbb55d70cde2f3a9544bf11","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-21 15:15:05.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af605f277565c2f6c9a6cc39c87bf1aeccff8297","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-02 21:23:25.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/916788/4/20250303045410/NEC-ISCSI : FAILURE in 54m 23s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/916788/4/20250303045410/NEC-FC : FAILURE in 1h 29m 12s","accounts_in_message":[],"_revision_number":4},{"id":"60c84a35c43b0e122d1027b0cd2ac00965616611","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-10 15:29:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5c355bc94aec025c4ce4ea18eff5b717af291007","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 15:30:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c94b8d6a86de18cbc6d1b1c52e868130d9bde33e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-10 17:26:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cb3ee14f1b502cc810bfc802e04fa77be42a4c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 17:27:38.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7042b38697cf4bc38fcbba9ad37b7e78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f2f6e1be7447299f75bc049b83d72f : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcc2614a2b6e4d2aaf08873ebb58fda7 : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fed57baf0d1a4d73b2a3c7f178884cfc : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d23270bc694960ad61cee114d0450a : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/d729732bfa584a4d8327946749c6c0c1 : SUCCESS in 56m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ca14b8922f2408f8a3d3670c56466e0 : SUCCESS in 27m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/868b6507aace4f488e2a712ba5d608a7 : SUCCESS in 1h 29m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49506f1c7fcb452c9f77d2a3a8f07a6b : SUCCESS in 53m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4cdd36d00d6b4655b0800dd39be69a80 : FAILURE in 52m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ca722bf84d7f4a08b559d3440d5f3ff1 : SUCCESS in 1h 55m 07s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/005cf5026bc742d5b0049c3126b45140 : SUCCESS in 1h 12m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6dd2fd4ee5e4cca849f0933456e7811 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":4},{"id":"b26785e39218167a8e3aeb1fd2543a1aba02a560","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-04-11 12:31:26.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"40e2d60b0ff120998bd507ddca072429eaa07975","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-11 15:41:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8689fffde6f21754b18b86d292358b0c9cf74e09","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-11 15:42:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eb8b5efbcbf5c5dd7542df2a0dd60961ab981495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 16:08:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43fd2315860045938c15bce86de2bd6b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e327ac7318f14ec09316622a811e9d68 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0dacc574f6c04b90a27a032864ebc6e3 : SUCCESS in 16m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c310e1a7a4634a16850c340b2c45be97 : SUCCESS in 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cf422e0b4c9d5a3dba133253954c1e48f0d0a73b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-11 17:29:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/26e537a2383648c29d4a56138bc2d051\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/41a114eb61ae4467b41eb52984074a68 : SUCCESS in 41m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/91b2d3386693480e8c2fe0aa2d84e2ac : SUCCESS in 43m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/75c07264783744b5b57a6610e727e590 : FAILURE in 1h 47m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e44dc98133514a68bfce94963a3d4244 : SUCCESS in 40m 22s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f86ccec75d1241ca90b23d42297a2eed : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d7c7d171a1564c4896ea748a7ae70b22 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bb67043087354c66b3bc0cd05350b26c : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/52ad6baa9f9e4627a76b1fa3e47d1ce4 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"1ba78e58c6b30c405968ca281b80efa695c99c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 18:18:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceec0bc979a44457b5270e27322547d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e5d4938912488aa2f044cef9b7a77a : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55dabf1a6b1a45a4b5d9ea6f1d53fb05 : SUCCESS in 13m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10d5a902bb6848c7aaafd9617c876670 : SUCCESS in 13m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ac8fb7c065c4fe3a417404a62ca9970 : SUCCESS in 14m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbdac048ce60475586378f53a63352d9 : SUCCESS in 14m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/498864ce58cb4754ac35ab1a93aa1ac0 : SUCCESS in 54m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af59d1eee6134a2680bb97d40dcbfeff : SUCCESS in 53m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb9a772ffe0746519a7263d2086f11a7 : SUCCESS in 1h 20m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c1bffdd090a484c8a47b454715b2dd0 : SUCCESS in 1h 07m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/55d260f6c0eb48489f66299009f15675 : SUCCESS in 31m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ae47c4fda0d24d2aa027d6ce8afb5419 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/feb334af11fb43a18418cbe6ab4313bf : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3030cbd8994844559c1d93d1493ba647 : SUCCESS in 9m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2a55de75bdaa4cf8b327a9b35ccfc60c : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c09ae72079074200811cb4700f65ba29 : FAILURE in 31m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ebd793f69fb84a74a7b2cf8c3230c347 : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ead9b2d5cb14a3da185a456268cab49 : SUCCESS in 2h 03m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8fd5af3512434c0dafef271091f07e13 : FAILURE in 2h 30m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d89d1b3acd041d88df38a3dfd12ac4d : SUCCESS in 1h 54m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/03fdf214dfc442d18a5157c1b2759f97 : SUCCESS in 2h 10m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/33a8a2535ad146bea7bde0694cdac477 : SUCCESS in 39m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2fcd2fcf1f0b44ed91fccb40d75a5205 : SUCCESS in 1h 09m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/792203c7182f4fe0a138ac7b092eafd5 : SUCCESS in 1h 57m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/396e7efbebf2420d9dd764da46877eb2 : SUCCESS in 40m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b7869e09a8a84350abe4ada7872c360e : SUCCESS in 1h 42m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/58b25c40f41e41e9ad44f891d3e1f610 : SUCCESS in 2h 13m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a39d40eb560c47e194478365c05503ba : SUCCESS in 1h 26m 48s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71cec84226234f2da2f6876c5d5ed385 : SUCCESS in 1h 15m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2ad5339bd747bb9f98100342c52477 : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":4},{"id":"b5a8253980814ce9b236ccbbf4889c992c98b620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 18:19:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fcbe9e4bba46e8b7530f94b723720c2b7bc09e34","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-11 20:17:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/24723e2d3e95431398c794ab268836b9 : SUCCESS in 2h 20m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/41f4285f4221412bb636c48938959ecc : SUCCESS in 2h 22m 28s","accounts_in_message":[],"_revision_number":4},{"id":"c33cf7ba16baf758d0b7367c7698c8e6725aac82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 20:44:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a9e2e2bc0d947b09aaab45cd6e8e787\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3abeff819eb24099b8d17653093d4a0f : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20d72bc1d8cf4ec09e28fc28bc8f3d73 : SUCCESS in 10m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20588e2926e94ac1854419c25a17fe9c : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a416f6a6c643f6ab1c63044a8a9346 : SUCCESS in 16m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/ae962fcfc6b640d389f90ad628264633 : SUCCESS in 1h 07m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6a81a56f28244e49f0ba75d0497c486 : SUCCESS in 55m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/912b1ca580014f5ab262a907e45d2a14 : SUCCESS in 1h 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/777c29981f0e467fa2db517102808252 : SUCCESS in 57m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0181a4a9f37b403891df0534931372a7 : SUCCESS in 1h 08m 21s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8d859a030554492a3ec5e105d73a371 : SUCCESS in 2h 20m 22s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e95746c2a2d645b9befe9ff97e334eb0 : SUCCESS in 1h 25m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a4111d76c6b4b2792729d8d08714f3d : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":4},{"id":"46c99d41770d38fcad790c3074b15e71f00bc2be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 20:44:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ad9c16800fda995237a6d8a72b43333137713600","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 20:46:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb5bafefab4f47a3bba7effb3c41503d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a720c4b4c1574ad3a086777f0dd6711b : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":4},{"id":"a58ca9b57993a615b573541cded422c2c1bf8925","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-11 21:19:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3e14379c28402aa350a033c5e4bb8954ee5ba256","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-04-11 23:21:37.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"717cd3d1ec8a4795cf7d29053ea8e720b32bd262","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-04-11 23:37:47.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"0705fe688fda17d2bd916e616d759cf781189ad1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-04-11 23:38:38.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"0e502690e8764e639cb4b88bd5ae1f37f2eb9a3b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-11 23:54:43.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"ae31150066ec6804596073936e7df9330c443e2f","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-04-12 02:56:01.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/e5fbc94fd52b40bd85bb6ff734d6bb6e\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/52e014ae770c4000afc3b9e94fd8917b : FAILURE in 3h 01m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4c0d0f8e986cf7c78f083f3df88bc13db99ca044","revisions":{"a1925405aca2eb4bc313611ef7a0ee20c34123b8":{"kind":"REWORK","_number":1,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/916788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/916788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/916788/1"}}},"commit":{"parents":[{"commit":"714ceff04623c8c1bb7095990dbc4fab24d98927","subject":"trivial: Remove trailing spaces, mixed tabs/spaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/714ceff04623c8c1bb7095990dbc4fab24d98927"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:41:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:21:06.000000000","tz":60},"subject":"trivial: Prepare for flake8 bump","message":"trivial: Prepare for flake8 bump\n\nThe version of hacking/flake8 we are currently using does not support\nPython 3.12, which means the \u0027pep8\u0027 tox target is not usable without\nmodifications on systems where this is the default python executable\n(like Fedora 39). Prepare for a bump by fixing all the issues seen with\nthe latest hacking version, 6.1.0. These mainly fall into one of three\ncategories: using keywords like functions (\u0027return(foo)\u0027 instead of\n\u0027return foo\u0027), missing spaces before trailing slashes, and use of \u0027\u003d\u003d\u0027\ninstead of \u0027is\u0027 for type equality checks.\n\nChange-Id: I0164582af81daf49b7391944940d2f8c187a9d42\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1925405aca2eb4bc313611ef7a0ee20c34123b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1925405aca2eb4bc313611ef7a0ee20c34123b8"}]},"branch":"refs/heads/master"},"cf9a45338658c0e301d265c47acb5b305d966191":{"kind":"REWORK","_number":2,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/916788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/916788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/916788/2"}}},"commit":{"parents":[{"commit":"c32535d29503b8756149adb4f654780c4ee4e1e5","subject":"trivial: Remove trailing spaces, mixed tabs/spaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c32535d29503b8756149adb4f654780c4ee4e1e5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:41:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"trivial: Enable E275","message":"trivial: Enable E275\n\nChange I4657d46d27ecfc45299d398cd2f3848fbc64b5b5 skipped this. Unskip it\nnow.\n\nChange-Id: I0164582af81daf49b7391944940d2f8c187a9d42\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf9a45338658c0e301d265c47acb5b305d966191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf9a45338658c0e301d265c47acb5b305d966191"}]},"branch":"refs/heads/master"},"7b078e84cbfbc102d2d384eafe449936de458da1":{"kind":"REWORK","_number":3,"created":"2024-05-18 02:53:42.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/88/916788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/916788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/916788/3"}}},"commit":{"parents":[{"commit":"6199583afe40d6b79dda564e0982be0a14740a3e","subject":"Merge \"Remove SQLAlchemy tips jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6199583afe40d6b79dda564e0982be0a14740a3e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:41:20.000000000","tz":60},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2024-05-18 02:53:25.000000000","tz":-300},"subject":"trivial: Enable E275 harder","message":"trivial: Enable E275 harder\n\nChange I4657d46d27ecfc45299d398cd2f3848fbc64b5b5 skipped this.\nChange Ib9f9a8bc6720ca8e1b916b2fa1583982859b08ae unskipped it.\nBut some things fell out while we were shaking the code.\n\nChange-Id: I0164582af81daf49b7391944940d2f8c187a9d42\nWas-Signed-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nRebased-by: Pete Zaitcev \u003czaitcev@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b078e84cbfbc102d2d384eafe449936de458da1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b078e84cbfbc102d2d384eafe449936de458da1"}]},"branch":"refs/heads/master"},"4c0d0f8e986cf7c78f083f3df88bc13db99ca044":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/916788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/916788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/916788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/916788/4"}}},"commit":{"parents":[{"commit":"a14312cc34e49d9623846548719ec5472463b1f7","subject":"Merge \"Add cinder-manage command to update service_uuid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a14312cc34e49d9623846548719ec5472463b1f7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:41:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 12:52:28.000000000","tz":60},"subject":"trivial: Enable E275 harder","message":"trivial: Enable E275 harder\n\nChange I4657d46d27ecfc45299d398cd2f3848fbc64b5b5 skipped this.\nChange Ib9f9a8bc6720ca8e1b916b2fa1583982859b08ae unskipped it.\nBut some things fell out while we were shaking the code.\n\nChange-Id: I0164582af81daf49b7391944940d2f8c187a9d42\nWas-Signed-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nRebased-by: Pete Zaitcev \u003czaitcev@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c0d0f8e986cf7c78f083f3df88bc13db99ca044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c0d0f8e986cf7c78f083f3df88bc13db99ca044"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
