)]}'
{"id":"openstack%2Fcinder~916789","triplet_id":"openstack%2Fcinder~master~I7dd14b4087d92f6bdae575581e03f991c519d57a","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-10-24 16:20:15.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-10-24 13:10:30.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-30 20:16:18.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-04-24 19:51:10.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-10-30 20:16:18.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-10-30 20:16:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7dd14b4087d92f6bdae575581e03f991c519d57a","subject":"Integrate \u0027pre-commit\u0027","status":"MERGED","created":"2024-04-23 12:51:48.000000000","updated":"2025-10-30 20:18:47.000000000","submitted":"2025-10-30 20:16:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916789-openapi","meta_rev_id":"f57a233d5d001e1d8a722aa1aa8d2b2c08371a6f","_number":916789,"virtual_id_number":916789,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-30 20:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-24 16:20:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-24 13:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-24 16:20:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"CC":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-23 13:25:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-23 15:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-23 17:42:32.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-23 18:37:55.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 11:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-04-24 15:38:10.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 16:17:42.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-04-25 01:07:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-25 13:43:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-25 21:12:10.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-26 18:43:19.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-26 18:57:58.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-26 20:23:04.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-09 22:58:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-20 11:18:23.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-26 04:07:38.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-28 04:35:42.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-09-18 14:15:04.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-18 16:00:13.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-18 17:39:38.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-18 22:14:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-13 12:59:44.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-02-21 16:52:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-03-03 08:11:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-03 08:11:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-04-10 15:35:59.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-10-24 13:48:47.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-24 15:06:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2025-10-24 16:20:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-10-30 14:58:38.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"}],"messages":[{"id":"28397e5e837bbc3d074e5e3843d81cc54ecb76fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-23 12:51:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf593b232805b6a42e73ea0375bf8fd25d4555e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:25:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f920f486616d406fa1e2deb7bcefcc11\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89c03cab788b4cfba0a41c7133a88226 : SUCCESS in 29m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0870a225dd484d64a96a4c6e9689bb12 : SUCCESS in 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13f63c81da529f5947d78f65e0d6d3a46830bc46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:39:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b47e8c437a34ecabed5753a5b821314\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6a35f2afd5af41229454ec97b118f0ba : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5aa6b28e4541429a1daa7756d2f288 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2180859aa564ea3b4a398f554a0ce3c : SUCCESS in 10m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82a8fcc00546418ebfa66655b6db7d94 : SUCCESS in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/520a073c0a5847db87d3d055700c0d1c : FAILURE in 12m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2238180327334a6fb488b99ab783d3ab : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/753da32db3d84ffbb501f37d62b68260 : SUCCESS in 1h 00m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47350113c49349a7bfb6785135aa84f5 : SUCCESS in 1h 43m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03ebbb880d424beea01911132ae09edf : SUCCESS in 54m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/220216f098c34f47a0f54d64cd27fab7 : SUCCESS in 15m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e84cbac8ffbf44119b56bb7684df1b1c : SUCCESS in 17m 03s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a35fad3e1a6f485d96661d261843db09 : FAILURE in 14m 27s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fe30dd134f8143ba8a721b14ce4dd106 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03ff1fceb73942569a11272a2b74029f : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/171e843730ac410f82dbb0c6027521c2 : SUCCESS in 6m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5dc6b79a81954d5ca95438edb448dcbd : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/22a0f8a83b0947519efc4c4bf23129e8 : SUCCESS in 39m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e6dbd8021d874378affe8425507a0891 : SUCCESS in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b35fb6e666e445ab897ae5edf6454b00 : SUCCESS in 2h 07m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ffa6875ecf734d58ba19e72a0e41643b : FAILURE in 2h 09m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cc5847ed383745aea97b5a502e18b747 : SUCCESS in 1h 43m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/083765891e9245ebbf8ff3b92b81e069 : SUCCESS in 1h 56m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/564f0424cae6477c947e6640774f896c : SUCCESS in 40m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/878a280dfa574a9f8d9f193cdf3fb82e : SUCCESS in 1h 07m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b690110cccb4557abe2fb17b6fce71c : FAILURE in 1h 30m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0924e8f2ce4c4976bde160455f961fa0 : SUCCESS in 34m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/71d0ac4abcd54af5804e28efbd618e9a : SUCCESS in 1h 31m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f9953715904c481f9461c0bfcf20e001 : SUCCESS in 1h 14m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6c2d757a9d7404ba463334e9199943f : SUCCESS in 1h 46m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ce2fc61af644ca9ba9fc494d723c28c : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a4afc11052535a219f97b6e723ef68d587c5dfd9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 17:42:32.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f56a519e85d14e8799160daf5d07a397\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0783f8b59d334195b3034872fe6bb0ef : RETRY_LIMIT in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"49410e887814bec9f6798ea082e53c305bcbda82","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-23 18:37:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916789 : FAILURE in 13m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d20772f096130a420a56effd9e13ffe30a4f6895","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-24 11:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/916789/1/fc : SUCCESS in 0h 51m 21s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/916789/1/iscsi : SUCCESS in 0h 59m 06s","accounts_in_message":[],"_revision_number":1},{"id":"799cabb84e68fef942af20ca1bc9b204e7c41b1b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-24 13:14:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a14d5c2d35503107e3feb8f9573503ddcfcd2e8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:45:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb11632c1e045d8a64e20c085a420e9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cba4382ea7c44b49ac5e38d10911d1ba : FAILURE in 24m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2014a482510f4bfbb61e03f4b4837faf : FAILURE in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f4ff384f7617a6c24b3287452fbda7c1d37ac19","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-04-24 15:38:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/89/916789/1/check/fujitsu-eternus-dx-iscsi/2cf60d3 : SUCCESS in 3h 20m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/89/916789/1/check/fujitsu-eternus-dx-fc/86fb33b : FAILURE in 3h 23m 04s","accounts_in_message":[],"_revision_number":1},{"id":"299a5c8f99c38423ed6cc2c7bc98b79e5dbcadc4","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-24 16:17:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a9bd9bee261ce4b169087ecc09eb5b2ee68d5c7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-24 16:23:20.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1c1dd5131a814734bbb691ca802c9610\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ac2d37b6343d48ad89ede4079ed04cc5 : FAILURE in 1h 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a2d463bdaaf09cd3e1269e3bf729e9c81c39c819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 17:05:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da30b3cebe204350b272e9199eb0e740\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3b3a49736d724c6ebf043859877b32da : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27fb368bbd6419e915de6ea9947aeaf : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f3b19d07c304fb4a159f506dbc44d53 : SUCCESS in 11m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2d3154031474ac2876667a3fe81b085 : SUCCESS in 10m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e85aed04293c467aafb672191fd726d7 : FAILURE in 11m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c10c0533a9a4082b8b21249241174e9 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/7d53d378b75b427598f6ba3e5ca18785 : SUCCESS in 1h 05m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fea45a666bda4daeabf27fd8343f5bf1 : SUCCESS in 1h 22m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b0ade9073c249b28fe00aca35a6c321 : SUCCESS in 1h 04m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f22b1b25b89241f1aa70623bca18c8cb : SUCCESS in 15m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/55dc8ea8fbb94204b30eeaf5c3408120 : SUCCESS in 16m 02s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/309aac23345043d8b2a7728d1c0ab065 : SUCCESS in 14m 44s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9e78641113d74ffa9a1ef1e27bf49254 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8144f5efe064a5f988348198e159afb : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5955e2a2848d4353ae669b9ce8ba4f73 : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4288622b03384832b29ed82b4af24735 : SUCCESS in 7m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c9872d7f07a64cd59fb70ed322e7fe8e : SUCCESS in 43m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff3f4b5df0da4b6ea7bcea0d359d1105 : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c363831e799b43beae0f92fde0c45eb9 : SUCCESS in 2h 16m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d10e78df2d76447296f4630801ab95af : SUCCESS in 2h 01m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/09fcd6bf307c4df6b18075aa4b7ee8e7 : FAILURE in 1h 51m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0f9dc917e73d4c4587cadf328dcbe99b : SUCCESS in 1h 49m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/64e37c0f829445888fe9f3522b2df89c : SUCCESS in 38m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ede117d33923425a8cb98e405e276ce5 : SUCCESS in 1h 19m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a5f1904b07c4032806ddb5e66cf1763 : SUCCESS in 1h 52m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e6673d7e24dc45788bb6709c6f344423 : SUCCESS in 31m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/811b1dde73924c57a5650aef345e8cdd : SUCCESS in 1h 27m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0bc3a61d5d6446a3be9ca4cdfae0fe31 : SUCCESS in 1h 06m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/04177f412294409f91a0580cbd961b50 : SUCCESS in 1h 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa76902058fc4ffea51867cf03714963 : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"42efd023059914fc6e02c8791d3a28bb2a6be99a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-25 01:07:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_Unity_FC/1e0ab5b/DellEMC_Unity_FC/1758 : FAILURE in 1h 29m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_Unity_iSCSI/8645798/DellEMC_Unity_iSCSI/1788 : FAILURE in 1h 49m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d4b05c6c69ead7f97a6b3a8b86a169197a72a4ea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-25 09:26:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01e8e7ada01f81e6f03fc7ed7942318376bbc238","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-25 09:32:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"755fa2411bfd665738f45bc5f0a6804d7e136b32","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-25 13:43:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/298/.","accounts_in_message":[],"_revision_number":1},{"id":"0b1df1694efd9a5daeae4b9cb895e6d852c135a8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-25 21:12:10.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-916789-1/FC/ : FAILURE in 40h 14m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-916789-1/iSCSI/ : FAILURE in 39h 6m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"999e9c5726ccb7052b2296fa24651ded1fd66166","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-26 16:03:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1853e7669f612b8ee572c6935dc169c42a075c62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:38:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2c369a8e32a4249bec796b2cf2908ec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc2c85a26e684426b4b84f7331ace90f : SUCCESS in 20m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c693923ffea541cfaa54861c15a9b8a4 : SUCCESS in 16m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4969e4eba2b64075157af83450c8049d0d1719b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 18:36:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc2b59baa206429980f5d3ae32f05ae5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a9a9247b858145808a2b025044fb2a83 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419f2c0ebfa9472c855ef1862e82c5be : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/717c01b535ac4a6b8766c12e03c7b81b : SUCCESS in 12m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aac09dec1c0e48518784502b8cc36fab : SUCCESS in 12m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adbcd9d749a1439c9d6eca4b26ac3598 : FAILURE in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7826f6b8a3e4262bf107bfdf5fd4617 : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/0a05a02aef1c4ed39c25f02e2145ae74 : SUCCESS in 58m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2262a48b790545cfb69249470cc24c2f : SUCCESS in 1h 37m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49ecb90f84ab4775af56b4db523d9463 : SUCCESS in 1h 01m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f7789073e8f434ea2a57b5871b529e3 : SUCCESS in 15m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d4738bac79644ed19f72a9fb20845618 : SUCCESS in 14m 03s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e28a068cdf6f4b03863589dea76be8b6 : SUCCESS in 13m 24s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/854ff72726d445478b6894cce80e9ed8 : SUCCESS in 7m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84b9dd42d9f04995b8279b0e8a9db82e : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4aaf67f65b024627add8a586a605cf43 : SUCCESS in 6m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8d542eae846f422898381cc9672469cc : SUCCESS in 7m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/393f59998e9d4c9faf2b21fd5ae91f61 : SUCCESS in 39m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a891c08c5c143afa0b70245594ea0a5 : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e2503b421a8d4ff392d7a8df6c69ad2c : SUCCESS in 2h 15m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/448a69ac4a114fbb889c5ce8e44423ee : FAILURE in 2h 15m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/37784826d4ce4887bc98041dc39d7a3a : SUCCESS in 1h 54m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c5dc0aaec5fb4e9195437ac45df7ebeb : SUCCESS in 1h 53m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6bb7eed659b640be947bf399a09ef395 : SUCCESS in 32m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/29aba9ff5904490ba7376d1aecbc0ace : SUCCESS in 1h 05m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7c0ec1b84ccc46309fc5c22e43764afa : SUCCESS in 1h 42m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fcdc99040a9d41d2bc5a3230b501a993 : SUCCESS in 54m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/95320657965d405dada2e1307d1b6e74 : SUCCESS in 1h 20m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bd485a5721c14f41a066e2d4b765c093 : POST_FAILURE in 2h 06m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6baa3f31fcf14749aff5e751a5617113 : SUCCESS in 1h 29m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb19194fc24c4fa49b210c186fb160f2 : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d78b6cd1623f5c638993792e681d614a836d253e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-26 18:43:19.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_04_2024_18_37_04\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"f7f7ca7cec6b4266699f3583bb9eb1e5fd9044a9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-26 18:57:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/916789/1/upstream-check/cinder-tempest-plugin-ontap-fcp/7740cdf/ : RETRY_LIMIT in 3m 40s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/916789/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/4f5159a/ : RETRY_LIMIT in 4m 21s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/916789/1/upstream-check/cinder-tempest-plugin-ontap-nfs/93c143e/ : RETRY_LIMIT in 4m 01s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/916789/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d484db7/ : RETRY_LIMIT in 3m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"61f1b98f0fb0f5c5e6cba78c07a52029da8332e3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-26 20:23:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_PowerStore_iSCSI/9710a8f/DellEMC_PowerStore_iSCSI/1281 : FAILURE in 1h 15m 03s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_PowerStore_FC/71bb1d9/DellEMC_PowerStore_FC/1452 : SUCCESS in 58m 53s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_PowerStore_NFS/162c346/DellEMC_PowerStore_NFS/971 : SUCCESS in 1h 04m 03s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_PowerStore_NVMeTCP/3fbaef9/DellEMC_PowerStore_NVMeTCP/832 : SUCCESS in 1h 45m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe239aa9c037872d813e4c28c11365f22b24f05a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-26 20:39:00.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_Unity_FC/282de54/DellEMC_Unity_FC/1783 : FAILURE in 1h 40m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_Unity_iSCSI/579db8a/DellEMC_Unity_iSCSI/1811 : FAILURE in 2h 01m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b1a856d9bac4821f57e08db6a4fe7f372801cd35","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-27 02:12:54.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/be30de73979545bcb4c3c2fa9bf8e8e4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d88828b6ba3844e3bf264751890dbc04 : FAILURE in 1h 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"88c1aadd05cff972db8b21b51d734204a4cd3073","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 22:58:57.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_PowerFlex_3.6/02b04a0/DellEMC_PowerFlex_3.6/1260 : SUCCESS in 1h 50m 24s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_PowerFlex_4.5/67039b6/DellEMC_PowerFlex_4.5/196 : SUCCESS in 2h 03m 49s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/89/916789/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/ffe0d3d/DellEMC_PowerFlex_4.x_NVMeTCP/136 : FAILURE in 2h 54m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"3fb136e5a7a3f73d2c10dab48022bc02786e2cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-13 14:51:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8d5b084889afc3d2af96c04fc7e7db421b83ae19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:15:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f0da55a85e40d9a24239e03392b4da\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed47fc2aa42f4044a879b74ca6242484 : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b887002acc8040d88945504d1a6e6cee : SUCCESS in 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"61744a90541abb0d259d0809c5018553a4894481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:14:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bf3f8abb908486c80822e7fb8694fb6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/14925e3a59f44d028cee957c569fae88 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b597d8ccf9e4728bf402576b5a895c4 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77cfdd798e5b48bd85e61a9625c3d2f2 : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a97b7b82327345438638275c3fb2d9c0 : SUCCESS in 11m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3ddfdf5378141f180d43e333cacf00c : FAILURE in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9056c41cf4c431b986b1075590f0fdc : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/6e47796260e943068769d5f2644fe599 : SUCCESS in 58m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/360357f5de5243c8bd8fbf0c35012cb8 : SUCCESS in 1h 20m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46431e24db6e41e7a78fedbc045debda : SUCCESS in 50m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ada0a3e03de446cb0913f13a1dfd892 : SUCCESS in 14m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce3b749c891e4330bdc8c1c2ccf1bfcf : SUCCESS in 15m 40s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/78c9c74525af44b8a644498305a76bf5 : SUCCESS in 12m 01s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5a511ad155b94f71b3c3361ad9fd106f : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c66a991221e4b4dbfe62a728f3b3113 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/496c415998af4882b9e4796bbfd8df82 : SUCCESS in 5m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6b89b714d14c403ca94e48dfdb3483de : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/64416a2cbf8940309e4d52a8d0f94dba : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a7b3875b8e34400d8e3e7b4242c27835 : SUCCESS in 7m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/400a5dfad71b44feb130f7d25053cb04 : SUCCESS in 1h 51m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/90fc03b55fdd4b0db514186a81716a9a : FAILURE in 2h 00m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/932d11cdb628442390ed97de14a24628 : FAILURE in 2h 13m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7234196e68c444129f3239cfc2b48260 : SUCCESS in 1h 49m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3cf13c4d37c7481591b1105935b20382 : SUCCESS in 29m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e0221b11225c4230855132f0aa956652 : SUCCESS in 1h 06m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b4b73c06f42d43acaf35533130ed2dd2 : SUCCESS in 1h 25m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/95462b40372249f8945bc8a13a10f34f : SUCCESS in 37m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/84902758ef364a10815de196afc8178a : SUCCESS in 1h 23m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/290e25fd41a34cacb329b56d740af9ba : SUCCESS in 1h 05m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c752da8adc2d467e9a306172dc84bcea : SUCCESS in 1h 28m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18c5d5e26e3142a5b7dee20507172242 : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":2},{"id":"c33e8b284dae5db49374f9cebeccdfb6702aa084","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-13 17:33:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916789 : FAILURE in 48m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4a60400e9b5fa534ae1c273559d168724d22c4ae","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-13 18:54:49.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/99b6c3f86cce485ab9e8a3afe8be1139\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b47b59df975c458dbc7dc9fc868c4b67 : FAILURE in 1h 24m 42s","accounts_in_message":[],"_revision_number":2},{"id":"82b9886638bdfdde5ae390605c6954085b39274d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-13 20:02:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/364/.","accounts_in_message":[],"_revision_number":2},{"id":"73b4ba7eb87a8074ccfcc5614898d39bc2db8b22","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-13 20:16:43.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-916789-2/FC/ : FAILURE in 4h 24m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-916789-2/iSCSI/ : FAILURE in 4h 25m 9s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"e30fd559d709439da12bfce3c6f85659a12a4182","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-13 20:50:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/916789/2/check/DellEMC_Unity_FC/6fa76f0/DellEMC_Unity_FC/1843 : FAILURE in 1h 34m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/916789/2/check/DellEMC_Unity_iSCSI/cb83a00/DellEMC_Unity_iSCSI/1871 : FAILURE in 1h 55m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c7092e131f0a68445a4fc220c9f945428ea37394","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-14 10:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/916789/2/fc : SUCCESS in 0h 41m 02s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/916789/2/iscsi : SUCCESS in 0h 47m 49s","accounts_in_message":[],"_revision_number":2},{"id":"118defb4cf12e7acfd344501d4a0232339ea33af","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-14 20:58:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/89/916789/2/check/DellEMC_PowerFlex_3.6/052db12/DellEMC_PowerFlex_3.6/1325 : SUCCESS in 1h 39m 10s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/89/916789/2/check/DellEMC_PowerFlex_4.5/be895c2/DellEMC_PowerFlex_4.5/250 : SUCCESS in 2h 16m 32s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/89/916789/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/0de4bb4/DellEMC_PowerFlex_4.x_NVMeTCP/197 : FAILURE in 2h 55m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8bd895d8c2fdea30944f67f59896c73791ced0d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:00:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7e86323dc7acdad321f6fc94417b31c92d5f5941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 10:07:27.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5d8a868148c36d3af5eb8cfb33657677567da614","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:30:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2a1aa5d7284e5d9f77bf186789f666\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6bc440ddc214231bb83c5abd5cfc05a : SUCCESS in 18m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dab759c3f6fd49f6855689b9c45f968b : SUCCESS in 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f625c74053d3afb10915187923dec2c22ea01af8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 11:18:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916789 : FAILURE in 23m 16s","accounts_in_message":[],"_revision_number":4},{"id":"b2df5b34a8855a57d62938105387e146989854d4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 12:07:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916789 : SUCCESS in 43m 53s","accounts_in_message":[],"_revision_number":4},{"id":"b96c7c7177e34edadd8ca9e54ad185a997a469ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:23:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02bc92bffce745e09822688b9f7ed9df\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c1a56568c802445a8099b78903368854 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b1fd8aeed9416f9d6f0437f33695a2 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/becf154ffd464ab3acca9496ebf6edf2 : SUCCESS in 10m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bc22271a8ac4e41814af0577d081965 : SUCCESS in 11m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b630e43b0a740a49130c851109ca2d2 : FAILURE in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b308581b2e694cc589727c77ea742f31 : SUCCESS in 13m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/db4aad0e6d9f4fb39c8a5a1cd131654b : SUCCESS in 55m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/33a235ebef994cbca7bc7f17b0bc042c : FAILURE in 1h 16m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a31256ab0ee843058d5e0c87d458c4be : SUCCESS in 58m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/512fe4a6f856456aad0cac74a9671cf8 : SUCCESS in 12m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/35960ba186d743f0b5b2fb476d818ac1 : SUCCESS in 18m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e95738d0117040508a81500922ef8298 : SUCCESS in 5m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36e516aaeaad4102ad94a025acb88961 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cc43d264ea9e4be5a1b55a0aad79bc69 : SUCCESS in 5m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/77a7a735299b487eb64d208361a8c711 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f61b4a26dfb3431f8a00c1917b506c67 : FAILURE in 49m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42b4493a35054a2fb482cd0fc077e18b : FAILURE in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f46c98dced7348d4958b86c00a23a174 : SUCCESS in 1h 33m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e7d5c0a3aff04330af6b6189d30a168f : SUCCESS in 2h 12m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a680bd431aca441bac18d3576d99f3fe : FAILURE in 1h 40m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f06aeab940d44ee9ac879a7174b4a3c7 : SUCCESS in 1h 52m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e7bef7f1428b45a183254270107cfabb : SUCCESS in 26m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7048d95962d545c790ae2cffcf8b5767 : SUCCESS in 1h 17m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6dc96bfdeec045a4a9cd56411f299190 : SUCCESS in 1h 37m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/384d0163a89f436bb0027faefb1882ec : SUCCESS in 47m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/17069df4b47246f7a0d17eb043bafc9a : SUCCESS in 1h 16m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5a4f38b085b7452f90a62163c34ad95a : SUCCESS in 1h 59m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c244ad51705947ada6aa796901d303ea : SUCCESS in 1h 39m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b93d4164dc954630ba80c37ecd5d0442 : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":4},{"id":"0a64f4f3aaaf4fcc7ae4d48e75fd1a1d2bac504b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 13:18:24.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d30f020022ce4b5085344193999b6058\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9b1382491fb146cfa882330387a2f2a1 : POST_FAILURE in 9m 22s","accounts_in_message":[],"_revision_number":4},{"id":"c7a8e1f94baa35cf1dca746de040658da5ce66f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-20 15:01:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/426/.","accounts_in_message":[],"_revision_number":4},{"id":"7c72a4a86a7051beddabbea3956197ab84374f14","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-20 16:04:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"77006d0b1089c2303454c727d290f2920a81b961","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:21:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/545ebbe04e0d4087b4d103f565d81898\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a5e996b2f0043c4b10fb414dc4bcb93 : SUCCESS in 15m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e16476d80013498d82ee225a1b184b37 : SUCCESS in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"725d73b8ed2a2731116be68d7c9ab72d34afa52b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 17:21:58.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/227c2b2e35854fe18c870dbd2a31c175\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/93a65366796f45a4b0d1a49ab8bba526 : POST_FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":4},{"id":"89b237b719e7a290471afbd3f56fd8f017bbc641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:25:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/290dc40784784cafa482ac0db1cb0c0c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5f3f472d88c54f9187e5532d9b3194d6 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e698885aebf4d3886c4d8f0cb986f01 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e9f932c38b7480eb7e07cf36f41e570 : SUCCESS in 9m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd38ffb005284ad7aefa4f581c36b293 : SUCCESS in 12m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9e3acbd50ac4fdebf353ebeb2897b93 : FAILURE in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544fb15bab644b6e808d79c45eae7b22 : SUCCESS in 14m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c55238dbbf684607a536bb03b59aa05a : SUCCESS in 1h 07m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/230b58d2a8d24595aa8ff656fe6935e5 : FAILURE in 1h 33m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f12ef551bbf841f6a0776c7d1667a999 : SUCCESS in 1h 07m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f56430704741452c9edc0d55d59c32d3 : SUCCESS in 12m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fe604b63adfe40fda3ee9bc4d5405c26 : SUCCESS in 13m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e2195788eef44a5e81d15055d921d14e : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/39e5da8b61b34d6cb06646cb6022984f : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ab9da58d2e549b897a4a4e7045b5862 : SUCCESS in 6m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7c9821d77b44472ca94a575f6366d32c : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bdc0ec38243f4dc7a71b01fd5ca039ca : FAILURE in 45m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bb790b381b42425d80c24987105c1015 : FAILURE in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/05e196249de940be80177ed38d207ff1 : SUCCESS in 2h 01m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f91eab66f4a41629d6665af993e5dee : FAILURE in 2h 08m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/393092f4a0214edab028c86d0af6e0b7 : SUCCESS in 1h 38m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6b8a474cdcb9409e8f090e0ead4d1fdd : SUCCESS in 2h 01m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/db9971e800cc4a12ba182ee97aec75a7 : SUCCESS in 27m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c6ee4632fee349e2b0205777e052cc24 : SUCCESS in 59m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7fed64693edd4b779aa78c3133032903 : SUCCESS in 2h 14m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/87370d7520bf4c899779993a36c7921e : SUCCESS in 52m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4985a0558962414aa3178dfe5991d684 : SUCCESS in 1h 21m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/82a9cb83dd084a469a24ec3f9ec7a417 : SUCCESS in 1h 32m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a3dd0fb9a07745a29dc164175714f902 : SUCCESS in 1h 54m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77dc438aa5cd4f4db533bc0b96f59e96 : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":4},{"id":"3ff6a4bb3070d24fc8c4eb1426dd2dbc62babb1e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-20 19:07:59.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/916789/4/check/DellEMC_Unity_FC/f4e4a9d/DellEMC_Unity_FC/1903 : FAILURE in 1h 48m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/916789/4/check/DellEMC_Unity_iSCSI/97c9794/DellEMC_Unity_iSCSI/1931 : FAILURE in 2h 11m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"9897be0e1fe5e8215308bf2197c59b181226f769","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-20 22:43:09.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-916789-3/FC/ : FAILURE in 10h 21m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-916789-3/iSCSI/ : FAILURE in 4h 49m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"fda006b44589b8584261bb8ef305c48711402781","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-26 04:07:38.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/916789/4/20240526041843/NEC-ISCSI : SUCCESS in 4h 54m 13s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/916789/4/20240526041843/NEC-FC : FAILURE in 8h 48m 51s","accounts_in_message":[],"_revision_number":4},{"id":"bf3487dec49b8b0262e2915d3a3f7db48d45a765","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-28 04:35:42.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/916789/4/20240528052330/NEC-V-ISCSI : FAILURE in 7h 32m 24s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/916789/4/20240528052330/NEC-V-FC : FAILURE in 8h 12m 09s","accounts_in_message":[],"_revision_number":4},{"id":"cf0dcda2bf73c5da82c5d7cce95348ce0e2d40b2","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-29 08:00:06.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/916789/4/fc : SUCCESS in 0h 37m 42s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/916789/4/iscsi : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":4},{"id":"ecc4061ac9274a50044dcfc79e12376de05035a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-18 14:14:43.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"28017acc35254d68a852316f2894a02e7186664a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:04.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/581/","accounts_in_message":[],"_revision_number":5},{"id":"74a26c63b02bd034ae321c3e1c8099941ad1f1d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 14:15:05.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916789/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/916789/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"7ea5ffd43cda2744329da08dd2d83d9f2b2a1d08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 15:33:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab3fbeb444d748d2b507eee119589521\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ddbc3fe630834f23aecd670af9807433 : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/754a36c08de84ceea3ea58efa993cd44 : SUCCESS in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a9c16fc491934164099dad67d6b824841303611f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-18 16:00:13.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/326ed170781a4774b973f65bf2a75afd\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a21a0be0fbfd4007a49d2190ada0bdd4 : SUCCESS in 45m 34s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/30b58376f24a4a4089305aff9c94c450 : FAILURE in 1h 19m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de71421373c141c4bf9d837c1db68881 : FAILURE in 1h 21m 58s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/df55068af0e94777bb54fa4c7b546c58 : SUCCESS in 38m 30s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"b10a9d5a5e33692e9289839ea526a0851e3dcf21","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-18 17:39:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/916789 : FAILURE in 23m 29s","accounts_in_message":[],"_revision_number":5},{"id":"86017dc1fc6fda23a48e7d54af92735d33953f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 18:40:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb19f4f041740a58dc9669d6b39d9d4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c978430b0c704cd8a2a33b56d4bd70cd : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8cca0dced5e4c41b50cac9969bec7eb : SUCCESS in 18m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d43824f9f3d6412b92a030f071b3fbfa : SUCCESS in 11m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2d76f4967834704ac7fdea9e3cc1c2b : SUCCESS in 24m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a2ec5b52a784880b85731174e5ccff4 : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c5ac9ea652450586482cd5b0023a40 : SUCCESS in 18m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/5a0aee58af384ccbbe88ada8af3c699f : SUCCESS in 57m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acca8727c28247bf846a4a11c5ad38f5 : SUCCESS in 1h 23m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e7675cb43c347b283e071491093ad5f : SUCCESS in 28m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca93f474bd734cef8f7140943148c009 : SUCCESS in 20m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0c5636f561b74595a6aea9ca0ba33eac : SUCCESS in 19m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/232d68d1363845aa8e5d20b09dca7925 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c8b2ee95a2647b7983af2cc90d9e9fe : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e9607b6a78049e8a938d600d2eefd89 : SUCCESS in 6m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1809561f60aa40dfb774efa06eb08a2e : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/76726d46d1cc4a20a650f8419c4fb7ab : FAILURE in 58m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/526b5f3b5f02419c955390bcee2da996 : SUCCESS in 18m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4e5a75effb1c4d52b9b432a7bc8da15a : SUCCESS in 1h 46m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/760d1c1100144975b40d19253584177d : SUCCESS in 1h 29m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c412ae72d81447e8bd28e22a60280cd1 : SUCCESS in 1h 56m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/89de8cef358d401d8c788ce94b9eadf4 : SUCCESS in 1h 41m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f2b5836b3ff140a7b52345ae657683ef : SUCCESS in 45m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e6364a9e636b44a2ad92d5c27683e9a8 : SUCCESS in 1h 31m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7874919df8744ec6ae8549b04f501eab : SUCCESS in 1h 23m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/27e3bc8eb89d4357880150b29afcb558 : SUCCESS in 37m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9bc87174e1e247e19edc2f40dc0b8dfd : SUCCESS in 1h 29m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ac5ee223284c4c05b1fa48d23e262ef4 : SUCCESS in 2h 12m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8adcf675d0e8446394f468d296cbb53d : SUCCESS in 1h 51m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e9dc553c705442fbd451cacf1b88daa : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":5},{"id":"24b2fa242e2ff95403e89a565870384d312249e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-18 18:46:07.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2024_18_40_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"2142f0964e357402d6bdd3df4089947ae0cb60a2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-18 19:33:12.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916789 : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":5},{"id":"cb48f3468cb0932119b1e165943b75fd8865618f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 19:35:58.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/916789/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/0cc9597/ : RETRY_LIMIT in 8m 40s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/916789/5/upstream-check/cinder-tempest-plugin-ontap-nfs/09d82e1/ : RETRY_LIMIT in 8m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"05a638317d6bc4c5585828ac5c8314ef3c3b0d03","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-18 19:52:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/916789 : FAILURE in 28m 30s","accounts_in_message":[],"_revision_number":5},{"id":"ed27290538db6c8ce1a4c127f383e1d816f59ba3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-18 22:14:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e36653bd123d4da6b3218598ffa79e14 : SUCCESS in 1h 14m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0785586546e94d8fabcfe6b05aaca667 : FAILURE in 1h 19m 51s","accounts_in_message":[],"_revision_number":5},{"id":"9a1b1b4d86b3d071f3df43fefa27f62e2a71c224","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 10:58:07.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_Unity_FC/ea008b5/DellEMC_Unity_FC/2623 : ABORTED in 3h 00m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_Unity_iSCSI/f21dafe/DellEMC_Unity_iSCSI/2632 : SUCCESS in 1h 59m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e5cc7a2cc59d53414690c54ec007af962fb4f93f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-19 13:20:11.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_PowerStore_iSCSI/c819ffe/DellEMC_PowerStore_iSCSI/1755 : SUCCESS in 1h 19m 17s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_PowerStore_FC/f112148/DellEMC_PowerStore_FC/1953 : SUCCESS in 1h 02m 30s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_PowerStore_NFS/c22a36b/DellEMC_PowerStore_NFS/1448 : SUCCESS in 1h 18m 02s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_PowerStore_NVMeTCP/e45527d/DellEMC_PowerStore_NVMeTCP/1308 : SUCCESS in 2h 06m 40s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_PowerStore_NVMeFC/00d5e35/DellEMC_PowerStore_NVMeFC/272 : SUCCESS in 1h 14m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"bb4feea7e63c78120eaf215eac1558659840db09","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-19 17:10:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/556/.","accounts_in_message":[],"_revision_number":5},{"id":"1e1eada89d9fa5ab1ecaf14b806cfada3c61f474","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-19 22:53:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1144/.","accounts_in_message":[],"_revision_number":5},{"id":"dc7d24a65a5b2fc0aef7d0e841f35032bf2216e2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-20 06:36:08.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_PowerFlex_3.6/4d9b45d/DellEMC_PowerFlex_3.6/2117 : SUCCESS in 2h 39m 16s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_PowerFlex_4.x/021e4e3/DellEMC_PowerFlex_4.x/1001 : SUCCESS in 2h 03m 27s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/89/916789/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/7bc4388/DellEMC_PowerFlex_4.x_NVMeTCP/941 : FAILURE in 5h 53m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"841cc1a0d98185d76aae0ccb9990457c97e8e913","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-26 00:21:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/916789/5/20240926051535/NEC-V-ISCSI : SUCCESS in 3h 54m 38s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/916789/5/20240926051535/NEC-V-FC : SUCCESS in 4h 06m 07s","accounts_in_message":[],"_revision_number":5},{"id":"a2bea1dc9123865cf0f5aa56d4d10b235bd70537","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-13 12:59:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/916789-5/HPMSA-iSCSI/ : SUCCESS in 3:16:59\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/916789-5/HPMSA-FC/ : FAILURE in 44:01.48","accounts_in_message":[],"_revision_number":5},{"id":"73380f561445ed1792c715f21b139cae7ad922ae","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-21 16:52:19.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8618114a95faea4a480386a4ecfd26dfe0e9bc64","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-02 22:51:22.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/916789/5/20250303062329/NEC-ISCSI : FAILURE in 51m 33s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/916789/5/20250303062329/NEC-FC : FAILURE in 1h 27m 51s","accounts_in_message":[],"_revision_number":5},{"id":"216105366e5cffb573db3534255ddf62107fa21c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-03 08:11:07.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"05b29b6bb2b578f63c74c7b8e34ca3cefb341583","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-10 15:35:59.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d8a914a494bf583b40b21292c2f0b8563997b755","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-10 16:27:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d76fc9a0671b10a1a1921fce41afceb9b2b538c5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-24 19:51:10.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"ae9d4fe44aca592a3ef58ffb43bc687278da7395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:47:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"3cb695cafaf1fef28244c4b4012363f6f54c749b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 12:49:10.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"5ad5d9ed37fddba1eee5de1d5a5e0a106f593fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:50:36.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"576d8299534c9a97d94ea987ded4a0c15fa6b70f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 12:53:18.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"fdc9dae6f67d9452b4ef73e67da4d2f31bce2b13","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-24 13:10:30.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"044f248203a7ee0233f2be81f5457682cd4d0e5c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:22:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"fac897a983e16c9ee761d8738bba9340c7e8977e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:23:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/083a0115a84e46ea9fc582901f07fa03\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e1fd307389ae44af8986f6ac7f9bc003 : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3eb43e7f29224efdae8b7eadf7c93e1f : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2642eec3b29341c1a4d6cb43287f805c : SUCCESS in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1f8cfdefc112657e718f911424372cc09800efac","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-24 13:28:29.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a2fa84d0163b3dae96a4e9908eeb979e226dd9a8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 13:48:47.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7dd14b4087d92f6bdae575581e03f991c519d57a .","accounts_in_message":[],"_revision_number":6},{"id":"9a4cacda617cf6aa625a97a90537631fce704354","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 14:20:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2246/.","accounts_in_message":[],"_revision_number":6},{"id":"1f10f55f3d2b21a1f444763279ffe4039375595f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 14:32:31.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7dd14b4087d92f6bdae575581e03f991c519d57a .","accounts_in_message":[],"_revision_number":7},{"id":"1e0756c093a099cfb83c772515ced7f32472cc08","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 14:50:36.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7dd14b4087d92f6bdae575581e03f991c519d57a .","accounts_in_message":[],"_revision_number":8},{"id":"d4404d9983c4d9220b23bbefa6f03d9d29e99876","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 15:00:08.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/29d1c69ce7d24bfcadb875078a92908f : SUCCESS in 48m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/660d35359ee0401298262aa3377adc0a : FAILURE in 54m 06s","accounts_in_message":[],"_revision_number":8},{"id":"deb1cae567adfb0b1c88eb35aa8cdf2416e6789e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-24 15:06:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6482b6cfa51512aff6d780df917b1a449416b600","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 15:10:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2132/.","accounts_in_message":[],"_revision_number":6},{"id":"a64bbccda2ee84f83d69ed40abe25b0f7ff1f2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 16:02:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f505f15cfed4a0a986e021a51853d4e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/404fd671595c437882d18dc5a4118d20 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc731b1930354df1b428f3ddf45536b3 : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb6681b32371436393255d0a501c5c26 : SUCCESS in 13m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c54773ac59cb4551ac3cd6fb1ca53149 : SUCCESS in 16m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeed452977c8469bade6da74f059c44e : SUCCESS in 15m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/9c6bd3a539de42bc939a60504d2fbe54 : SUCCESS in 1h 09m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4821226c9967479a821d72589f837c9b : SUCCESS in 57m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df0cafb88d604287bd138df243ae83e9 : SUCCESS in 1h 28m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e0da509eafa4ee3b113d378850c0f1f : SUCCESS in 1h 02m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a20b418ff0c540faaaa740f8e6c17304 : SUCCESS in 18m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/36628a2133a0401a91f962350184bb28 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6808497dba0d42ea9ae8efb971a67d9c : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91bf7ba68a304412a2dbcaffe903626e : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c3b1969aa91743fd9467ebf789cf309e : SUCCESS in 6m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/88bb558236e84d4a9414ac9551121f8f : SUCCESS in 8m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a9ef72aa8924eec9242bdd1fbef5007 : FAILURE in 37m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef330bc8ea7a45408f9ff4d9ffc101e9 : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6d1d26dcc944bac8f23b019aea7e9bb : SUCCESS in 1h 49m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2bd1ac050d6d467b97624a2b53e2c195 : TIMED_OUT in 3h 06m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1aa684a1a4254b5bb2e091b1c2ecf91c : SUCCESS in 1h 54m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/aab31966624642b2a1b062c60e649632 : SUCCESS in 1h 51m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/517606458242451a8df8861f67332a5e : SUCCESS in 36m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3cd63122e8f04f5e8b395aef32ef9523 : FAILURE in 44m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b25fc49e8a8843518ef9bc5c124dcc15 : SUCCESS in 1h 37m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0b22fabc5d504ccc8bce672e81b3f227 : SUCCESS in 37m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/266aa9c7f5644687b2a80b6a076d4037 : SUCCESS in 1h 56m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/93dcc577ca3340d3a286f137537bb817 : SUCCESS in 2h 20m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e55131c7167f40f080d5290741471cb0 : FAILURE in 1h 49m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24dc5ebde60a48a799a5703ea287c3a3 : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":8},{"id":"a204d16cfe12bbdd1b7d8c45d1fe255329130f16","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-24 16:08:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1bf016491f8adea2c9c2d21b563b4545ac3e3df0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-24 16:20:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3a59bf10d8e64438d621271de93675cfeab496b1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-24 16:20:39.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"393a231baf505bc229902b900225c1180237bb3a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 18:12:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"22b430bb63559a583df4a2bf1c699df59a2428a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 19:01:02.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26e5b93dee604325bfd2237de53df390\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b1b6de8842654368a3b4c3cccefbb80e : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea37b3e9787d49f6a74008a46cde9036 : SUCCESS in 7m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a7fa093c6244996a81a1e2d56e9c77c : SUCCESS in 12m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4728043af3234f9b8173c11315519bcb : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c913ab31fab4fcd90cea8139a74e739 : SUCCESS in 14m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/31fe984615cd481b9b171f354b5538e9 : SUCCESS in 1h 02m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c1b89601cb844cbb4674c056e0eeef3 : SUCCESS in 1h 01m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed463a4fa02143a08f292c9fab947de6 : SUCCESS in 1h 50m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/818d60f5f7a948c6ba5e790598c73dde : SUCCESS in 1h 04m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cff1fc76ba94d5599d52f3ffcc9c28a : SUCCESS in 15m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5feacd6410bc4f39a57e6ffb3a6b8545 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2b487e8da5a54f9889ad6ee3e2142546 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cee5a1ac7314ef29d0c4c46da879b56 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/989668cd678e4524b7fae37f14e4081c : SUCCESS in 6m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5de82ded02a9488ab5046d6c08e74884 : SUCCESS in 10m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3f497b0d048c41eb92f6ce2d60603b10 : FAILURE in 26m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c4c4b2f840a74eb49b7d9c5ae548167b : SUCCESS in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bbf1ba57febc4a579845ae15034cedf1 : SUCCESS in 1h 59m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/865bfca8c98e4c8d98178c699b355497 : FAILURE in 2h 34m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b8dba08ab6d4459190e423569eaab6dc : SUCCESS in 2h 01m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/41b3c04cf99f4da6a1517f421d30e722 : SUCCESS in 2h 03m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ec2a0336a8a7424ab29c2bd4227d39c4 : SUCCESS in 19m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/12b3e9b0d98b476b9098d80589506b73 : SUCCESS in 1h 01m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4753ec5dfe0446a1a0c731c63468d7fc : SUCCESS in 1h 47m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4cda242312194445a2d3470da7a501b6 : SUCCESS in 34m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/273dc23d623b482d97e46331c56f38fb : POST_FAILURE in 2h 19m 00s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5fa89aeae6494b50b6b67f6939b96500 : SUCCESS in 1h 54m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90e6d01bd784463e88cf1d77da7fa459 : SUCCESS in 1h 47m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b6d8b6fc29b462f9cac64ad36dcbac8 : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":8},{"id":"ebbe3fe17fce3e2555db54ef8c42b5d0e4d8171c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 06:29:07.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2263/.","accounts_in_message":[],"_revision_number":7},{"id":"644185e704d3d0a39cb2b4b9003bc899b04a411d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 06:48:54.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2149/.","accounts_in_message":[],"_revision_number":7},{"id":"be8cb6cf9a4bfac73a659b701ee313ca122fb3d5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 07:22:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2264/.","accounts_in_message":[],"_revision_number":8},{"id":"c57a8cd2f83708798407537e2c7ea1922bfa93ab","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-25 07:40:02.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/916789 : FAILURE in 1h 11m 57s","accounts_in_message":[],"_revision_number":8},{"id":"5d588f1bf2df43a049157fcb6bf6596a31733b59","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-25 07:43:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2150/.","accounts_in_message":[],"_revision_number":8},{"id":"7a1f6793e20d3cef203ad8d4407310fda5a39c0c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-25 12:26:04.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b95879d25d67453d9fff223b26279356 : SUCCESS in 43m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/628baf2c05d841f0924314cae95cbfb8 : FAILURE in 1h 08m 57s","accounts_in_message":[],"_revision_number":8},{"id":"85fbda6e8eae023b126092199e26fa056f4bf159","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-30 14:58:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5c4a6d94c5c7b5b515ce8c028b85805c355da806","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-30 15:09:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"60578199a572c122183f4a53e62cca4ca37199eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 15:25:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1959c7ea0dd841cbaac4c1d80910c9e0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/266d7368210b4a3c80531cfddda6b214 : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6f663b810aa041d3ad4807c6422226a7 : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d2527b5b0c534eed9bfab55ff9cfb3a3 : SUCCESS in 16m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d0e4111deaae0d891c1fd040c5945e67a178f59c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-30 17:58:42.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/67e171c53f4b41ed90b5236697d31463 : SUCCESS in 49m 03s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/46dafda0e97d4bdd956d21923d4416d6 : FAILURE in 1h 01m 29s","accounts_in_message":[],"_revision_number":8},{"id":"573161cb6f7650a0d474d65c264ee12a0c9aa8aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:07:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40b31a6f18c343e0a29181b690be059c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b8cf6e451c1a4f3381bdc602cee8ec26 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de86195d96d34f53b972225b7936855f : SUCCESS in 8m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a13a5e76e804fa8a71b2e4f42037956 : SUCCESS in 12m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/127c8f293b19416aae8351043ed55364 : SUCCESS in 14m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9edee573395d4185a11c44ae24f5225b : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/d81bf3f241d6411286728b16e36cc2ea : SUCCESS in 53m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af6395f7006d41a1a4f10bc6880feba2 : SUCCESS in 33m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f29d34b42abe4d1ea26ab37b3cc1bf3c : SUCCESS in 1h 42m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13a955e5bf3f43ef9e7b7b8d265dec50 : SUCCESS in 1h 05m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dd2458ed2054b9b854ce0b583bddcdb : SUCCESS in 19m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b37409f81ad94d409a6cd0cbc60f966a : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/93c030f710ab408ea57bca96f5c52077 : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a1df7e65f864016b1778aadac42ec74 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14e78219fe81465da258dc9a0ab23692 : SUCCESS in 4m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d3e9fb42991547abb26608c42003ed7f : SUCCESS in 10m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7936ca26c3c47819f106472a8c06f2a : FAILURE in 31m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f089b2b2ce874576a75e87c274e67f1c : SUCCESS in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6e73f13462d24ad39822e98fcfbac2f4 : SUCCESS in 2h 12m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/343231cec8b64bd3a117d640ed11e8d4 : FAILURE in 2h 42m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f9b72ed6d824e67a7ad1a3f7ff73d67 : SUCCESS in 1h 47m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8c26b62843934e40ae0daafa3c54f1f1 : SUCCESS in 2h 20m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bb9781a7414047dc968c92713b8795aa : SUCCESS in 27m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1cd5ecac98044c98bbdf846422c09215 : SUCCESS in 1h 09m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6495e864a62d4531ab4648b7f87510e8 : SUCCESS in 1h 38m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/363bdf8cba2e4bf9bff86694310dbe61 : SUCCESS in 40m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/982a2c2e20f74e87bf7c2fa8cfdfbfd5 : SUCCESS in 2h 28m 37s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/806a633f3e4d474d99ec83158a0e9c83 : SUCCESS in 1h 22m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9553a75f86e543869a9ca7c1bf78a422 : SUCCESS in 1h 22m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94889a976bbe4b538ff13d4e90ddb847 : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":8},{"id":"2e8189d8da2d5983fafa2195e120df5b2b026a9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:08:28.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5cf487178a5cc5d31b3605f61486ccd01878abf4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-30 18:17:17.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_10_2025_18_10_43\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"2d3fc37a7c61fb3de6fb8ed7a468f10e04413c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:16:17.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d763a6191127473f848014faebb862fc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/efa9d7c12a83408ab7804b2722ba7284 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2597883012d46a39c8ab65d6cfc31fa : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d11659d659c94dc586879c0fef144ec0 : SUCCESS in 11m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/feb22888aca44410b9e1ded528d87d36 : SUCCESS in 15m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6bd71ffc3ed40ba91c8c4fe1ab0cd6a : SUCCESS in 14m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/8ba813426f484d52a2661e5b8a0e75a5 : SUCCESS in 49m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2593fe94e4f54cc781886593e60aafee : SUCCESS in 43m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bb6e8bd06e847be8f82120c3dc7f802 : SUCCESS in 1h 07m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15cc0cf949ef44b49fd610d700a873e5 : SUCCESS in 58m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c39e573122d480989e1808396b9610e : SUCCESS in 16m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3cb62f59f4a74b54aa77ec961cf25d7f : SUCCESS in 54m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bb4e089f7b084b87bf251274704a9c8f : SUCCESS in 2h 03m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e013dc41b6b64e359e0fc8622f2f8570 : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":8},{"id":"17ed8da306c8184ccc26825c585d1ea002de8e5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:16:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"f57a233d5d001e1d8a722aa1aa8d2b2c08371a6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:18:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae05710899b646d7a2a81f978ec3d6a6\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9c0224139fa04ccbaf31dc2ec7d56755 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c543f4f3f742169a403c2c4d68f873 : SUCCESS in 2m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90e52fcba74c4441a93890e78089d0e1 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"41eee5c2191ba6cb3af30ef2d2a93d1287df5ad7","revisions":{"44f07694582e11e98554c198228feebb91bca1c1":{"kind":"REWORK","_number":1,"created":"2024-04-23 12:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/916789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/916789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/916789/1"}}},"commit":{"parents":[{"commit":"a1925405aca2eb4bc313611ef7a0ee20c34123b8","subject":"trivial: Prepare for flake8 bump","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1925405aca2eb4bc313611ef7a0ee20c34123b8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 12:50:24.000000000","tz":60},"subject":"Integrate \u0027pre-commit\u0027","message":"Integrate \u0027pre-commit\u0027\n\nThis nifty tool provides a extensible way to configure pre-commit hooks\nto do everything from running linters to spellchecking [1] and has\nalready been adopted by a few OpenStack projects [2]. Let\u0027s copy what\nwe\u0027ve done elsewhere and migrate most of our linters to this. This makes\nmy life much easier as I work on the OpenAPI series since linting is\ntaken care of. Users who do not wish to use this can simply not do so.\n\n[1] https://pre-commit.com/\n[2] https://review.opendev.org/#/q/file:.pre-commit-config.yaml\n\nChange-Id: I7dd14b4087d92f6bdae575581e03f991c519d57a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44f07694582e11e98554c198228feebb91bca1c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44f07694582e11e98554c198228feebb91bca1c1"}]},"branch":"refs/heads/master"},"e9613b1bb23e3a834449a4a5074d783c061ab739":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-13 14:51:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/916789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/916789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/916789/2"}}},"commit":{"parents":[{"commit":"cf9a45338658c0e301d265c47acb5b305d966191","subject":"trivial: Enable E275","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf9a45338658c0e301d265c47acb5b305d966191"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-13 14:50:48.000000000","tz":60},"subject":"Integrate \u0027pre-commit\u0027","message":"Integrate \u0027pre-commit\u0027\n\nThis nifty tool provides a extensible way to configure pre-commit hooks\nto do everything from running linters to spellchecking [1] and has\nalready been adopted by a few OpenStack projects [2]. Let\u0027s copy what\nwe\u0027ve done elsewhere and migrate most of our linters to this. This makes\nmy life much easier as I work on the OpenAPI series since linting is\ntaken care of. Users who do not wish to use this can simply not do so.\n\n[1] https://pre-commit.com/\n[2] https://review.opendev.org/#/q/file:.pre-commit-config.yaml\n\nChange-Id: I7dd14b4087d92f6bdae575581e03f991c519d57a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9613b1bb23e3a834449a4a5074d783c061ab739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9613b1bb23e3a834449a4a5074d783c061ab739"}]},"branch":"refs/heads/master"},"51890c0d5162f407304c0e9e222ba20fac522224":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-20 10:00:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/916789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/916789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/916789/3"}}},"commit":{"parents":[{"commit":"240e6ffd03613d83e1881e74fb49a3717a9d8b66","subject":"trivial: Remove trailing spaces, mixed tabs/spaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/240e6ffd03613d83e1881e74fb49a3717a9d8b66"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 09:58:47.000000000","tz":60},"subject":"Integrate \u0027pre-commit\u0027","message":"Integrate \u0027pre-commit\u0027\n\nThis nifty tool provides a extensible way to configure pre-commit hooks\nto do everything from running linters to spellchecking [1] and has\nalready been adopted by a few OpenStack projects [2]. Let\u0027s copy what\nwe\u0027ve done elsewhere and migrate most of our linters to this. This makes\nmy life much easier as I work on the OpenAPI series since linting is\ntaken care of. Users who do not wish to use this can simply not do so.\n\n[1] https://pre-commit.com/\n[2] https://review.opendev.org/#/q/file:.pre-commit-config.yaml\n\nChange-Id: I7dd14b4087d92f6bdae575581e03f991c519d57a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51890c0d5162f407304c0e9e222ba20fac522224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51890c0d5162f407304c0e9e222ba20fac522224"}]},"branch":"refs/heads/master"},"33204d6f37b6d433099f0c3d98bc729c912a1e25":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-20 10:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/916789/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/916789/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/916789/4"}}},"commit":{"parents":[{"commit":"a75ed0b14973d78d4b5f92dbb2df18948ee47fd2","subject":"trivial: Remove trailing spaces, mixed tabs/spaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a75ed0b14973d78d4b5f92dbb2df18948ee47fd2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-05-20 10:06:23.000000000","tz":60},"subject":"Integrate \u0027pre-commit\u0027","message":"Integrate \u0027pre-commit\u0027\n\nThis nifty tool provides a extensible way to configure pre-commit hooks\nto do everything from running linters to spellchecking [1] and has\nalready been adopted by a few OpenStack projects [2]. Let\u0027s copy what\nwe\u0027ve done elsewhere and migrate most of our linters to this. This makes\nmy life much easier as I work on the OpenAPI series since linting is\ntaken care of. Users who do not wish to use this can simply not do so.\n\n[1] https://pre-commit.com/\n[2] https://review.opendev.org/#/q/file:.pre-commit-config.yaml\n\nChange-Id: I7dd14b4087d92f6bdae575581e03f991c519d57a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33204d6f37b6d433099f0c3d98bc729c912a1e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33204d6f37b6d433099f0c3d98bc729c912a1e25"}]},"branch":"refs/heads/master"},"841fad272c17fbf502e2c5d888d5914d1b65dd2a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-09-18 14:14:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/916789/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/916789/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/916789/5"}}},"commit":{"parents":[{"commit":"6b3b9ccdeb806df1114587d4666c73bffdaad67c","subject":"trivial: Remove trailing spaces, mixed tabs/spaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6b3b9ccdeb806df1114587d4666c73bffdaad67c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-18 12:52:28.000000000","tz":60},"subject":"Integrate \u0027pre-commit\u0027","message":"Integrate \u0027pre-commit\u0027\n\nThis nifty tool provides a extensible way to configure pre-commit hooks\nto do everything from running linters to spellchecking [1] and has\nalready been adopted by a few OpenStack projects [2]. Let\u0027s copy what\nwe\u0027ve done elsewhere and migrate most of our linters to this. This makes\nmy life much easier as I work on the OpenAPI series since linting is\ntaken care of. Users who do not wish to use this can simply not do so.\n\n[1] https://pre-commit.com/\n[2] https://review.opendev.org/#/q/file:.pre-commit-config.yaml\n\nChange-Id: I7dd14b4087d92f6bdae575581e03f991c519d57a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/841fad272c17fbf502e2c5d888d5914d1b65dd2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/841fad272c17fbf502e2c5d888d5914d1b65dd2a"}]},"branch":"refs/heads/master"},"60d30a37f7d0b8f717baa38f128cc6555e8ea4b1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-24 12:47:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/916789/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/916789/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/916789/6"}}},"commit":{"parents":[{"commit":"d02171164bdd702b12b59888b744d172f30d712d","subject":"Merge \"docs: starting iscsid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d02171164bdd702b12b59888b744d172f30d712d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:44:14.000000000","tz":60},"subject":"Integrate \u0027pre-commit\u0027","message":"Integrate \u0027pre-commit\u0027\n\nThis nifty tool provides a extensible way to configure pre-commit hooks\nto do everything from running linters to spellchecking [1] and has\nalready been adopted by a few OpenStack projects [2]. Let\u0027s copy what\nwe\u0027ve done elsewhere and migrate most of our linters to this. This makes\nmy life much easier as I work on the OpenAPI series since linting is\ntaken care of. Users who do not wish to use this can simply not do so.\n\n[1] https://pre-commit.com/\n[2] https://review.opendev.org/#/q/file:.pre-commit-config.yaml\n\nChange-Id: I7dd14b4087d92f6bdae575581e03f991c519d57a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60d30a37f7d0b8f717baa38f128cc6555e8ea4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60d30a37f7d0b8f717baa38f128cc6555e8ea4b1"}]},"branch":"refs/heads/master"},"68e0f8d690c4778dedbba90874b6dee0fc4051f6":{"kind":"NO_CHANGE","_number":7,"created":"2025-10-24 12:50:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/916789/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/916789/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/916789/7"}}},"commit":{"parents":[{"commit":"d02171164bdd702b12b59888b744d172f30d712d","subject":"Merge \"docs: starting iscsid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d02171164bdd702b12b59888b744d172f30d712d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"subject":"Integrate \u0027pre-commit\u0027","message":"Integrate \u0027pre-commit\u0027\n\nThis nifty tool provides a extensible way to configure pre-commit hooks\nto do everything from running linters to spellchecking [1] and has\nalready been adopted by a few OpenStack projects [2]. Let\u0027s copy what\nwe\u0027ve done elsewhere and migrate most of our linters to this. This makes\nmy life much easier as I work on the OpenAPI series since linting is\ntaken care of. Users who do not wish to use this can simply not do so.\n\n[1] https://pre-commit.com/\n[2] https://review.opendev.org/#/q/file:.pre-commit-config.yaml\n\nChange-Id: I7dd14b4087d92f6bdae575581e03f991c519d57a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68e0f8d690c4778dedbba90874b6dee0fc4051f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68e0f8d690c4778dedbba90874b6dee0fc4051f6"}]},"branch":"refs/heads/master"},"41eee5c2191ba6cb3af30ef2d2a93d1287df5ad7":{"kind":"REWORK","_number":8,"created":"2025-10-24 12:53:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/916789/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/916789/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/916789/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/916789/8"}}},"commit":{"parents":[{"commit":"d02171164bdd702b12b59888b744d172f30d712d","subject":"Merge \"docs: starting iscsid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d02171164bdd702b12b59888b744d172f30d712d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 11:34:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 12:52:33.000000000","tz":60},"subject":"Integrate \u0027pre-commit\u0027","message":"Integrate \u0027pre-commit\u0027\n\nThis nifty tool provides a extensible way to configure pre-commit hooks\nto do everything from running linters to spellchecking [1] and has\nalready been adopted by a few OpenStack projects [2]. Let\u0027s copy what\nwe\u0027ve done elsewhere and migrate most of our linters to this. This makes\nmy life much easier as I work on the OpenAPI series since linting is\ntaken care of. Users who do not wish to use this can simply not do so.\n\n[1] https://pre-commit.com/\n[2] https://review.opendev.org/#/q/file:.pre-commit-config.yaml\n\nChange-Id: I7dd14b4087d92f6bdae575581e03f991c519d57a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41eee5c2191ba6cb3af30ef2d2a93d1287df5ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41eee5c2191ba6cb3af30ef2d2a93d1287df5ad7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
