)]}'
{"id":"openstack%2Fcinder~917292","triplet_id":"openstack%2Fcinder~master~I0276f4fe42478670aa24f3e725adbbcab086082d","project":"openstack/cinder","branch":"master","attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-04-29 08:46:48.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0276f4fe42478670aa24f3e725adbbcab086082d","subject":"wip: Add type field to backup API","status":"NEW","created":"2024-04-29 08:21:53.000000000","updated":"2024-08-04 23:35:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32253385438be25cdfc665b919eecd796a32a6a0","_number":917292,"virtual_id_number":917292,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-29 10:51:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-29 08:39:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-29 08:46:48.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-04-29 09:46:31.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-29 09:56:45.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-29 10:46:04.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-29 10:51:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-29 11:25:37.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-30 11:17:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-11 18:28:26.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-04 23:35:20.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"}],"messages":[{"id":"a5a254b4191037b4390ff4799158fd60b1e0ba24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-29 08:21:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f1ed42b0c2c5741cffdc0b3222e4fd34f7cd547","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 08:39:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc3df877cc574c18ab3ce5e6e7888026\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04b93a11a2dc4e92b864e50b75145b84 : FAILURE in 15m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d6e636e2bf0431d92acbf10aaea126f : FAILURE in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2df128479a454cedf498a1c331e8c128f4061718","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-29 08:46:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/917292 : FAILURE in 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"15f021c790c422206112568920aef68dc4f3f45a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-29 09:14:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/917292 : FAILURE in 23m 45s","accounts_in_message":[],"_revision_number":1},{"id":"759ba08f517b60cdc0b4ee3f401df114d0e5e090","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-29 09:46:31.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/319/.","accounts_in_message":[],"_revision_number":1},{"id":"354931b44d224aba390e76cd0680dd396ac382ec","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-29 09:56:45.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9a745885452a45929d5726c02dfa69a9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a1df57e50f48411a8b8e9bbb6c00494a : FAILURE in 1h 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"659554e761b9c581808efe5dac8c886cc57d76ca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-04-29 10:46:04.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"daf178348f41a9ebf7a7eec95ac995a3376c40e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 10:51:38.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12c86f33ca284b9bbbe8d72e5db76689\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ecbd9f8a504e21a37436272ef7f25f : FAILURE in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61175cbf79e6419a919f3387222fad61 : FAILURE in 9m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6e9ddbfd17e47639ec520b490eac68a : FAILURE in 12m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4a2b9bef3564fc6aa2aacab8b9f2119 : FAILURE in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5cee7c61334d6f890ba83f334d9b95 : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/ea9f38340f924939923d139e4fec2b7b : SUCCESS in 58m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70b3797f57f44cffa526fbba9fb48645 : FAILURE in 1h 03m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b800434fd1fe47c195269cc2b4e6c0d5 : FAILURE in 1h 03m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7f00736a5c94362b5fba47b05f9ea89 : SUCCESS in 11m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1e6848e0aa9943eea21457a6d9d7d55b : FAILURE in 14m 41s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/176b618fc7f342569ca2b67feae0de71 : FAILURE in 16m 19s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ce5efd7d6d174c96b5ddeeb898544cb8 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd90e9075d6640549ae5e30c514f7289 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b604b720f3344f295fd4c1ec4a79233 : FAILURE in 6m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f650e497cd66454fb2e42c97a717c19a : FAILURE in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca8581625c43487bb2a2af80b4711e8c : SUCCESS in 34m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6029f2cf2d104395be7d867e7ace7256 : SUCCESS in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b2d78352e304340996c3989db49b0f1 : FAILURE in 1h 43m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/55b4b76ca24c407083e41cd41160229d : FAILURE in 2h 03m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65fd8c0ec7cc4a66b5ebb08f90a19209 : FAILURE in 1h 27m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/31a2b9c9e60744409e64839f18d070b3 : FAILURE in 1h 35m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1cf2f9fa913549e885aa218f060021f5 : FAILURE in 35m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ca7352053974ae493dc8d2816b7abfc : SUCCESS in 1h 10m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7d8bc5b4da0b4cb587e0198f2b2ebb27 : FAILURE in 1h 25m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a5efd1df29c245689b73bcade58e336d : FAILURE in 42m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3958ce291cbb4355939e766a7743bc65 : SUCCESS in 1h 18m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9a3d47809c8b4c83a0f4155259b2b3d5 : SUCCESS in 1h 38m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/851928f3378d498584f471529aa0ca17 : FAILURE in 1h 34m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e87b3289ce5f4f27a0f72de880758ba9 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fe56f24a0c071dfafe13050c4c2cc16ca8967c7a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-29 11:25:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/917292/1/check/DellEMC_Unity_FC/4b0fc73/DellEMC_Unity_FC/1792 : FAILURE in 1h 35m 40s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/917292/1/check/DellEMC_Unity_iSCSI/5b7de47/DellEMC_Unity_iSCSI/1820 : FAILURE in 1h 40m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"4fac39df2d21c82e9049d66180a635440e123ac6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-30 11:17:52.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/917292/1/20240430085831/NEC-ISCSI : FAILURE in 10h 44m 38s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/917292/1/20240430085831/NEC-FC : FAILURE in 11h 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e2e9dedb5f5a38107ecbebb6c82293ff965e3a52","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-11 18:28:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/92/917292/1/check/DellEMC_PowerFlex_3.6/88aa6d6/DellEMC_PowerFlex_3.6/1285 : FAILURE in 1h 44m 05s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/92/917292/1/check/DellEMC_PowerFlex_4.5/0b3b65e/DellEMC_PowerFlex_4.5/217 : FAILURE in 1h 57m 07s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/92/917292/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/004d115/DellEMC_PowerFlex_4.x_NVMeTCP/159 : FAILURE in 2h 56m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"32253385438be25cdfc665b919eecd796a32a6a0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-04 23:35:20.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/917292/1/20240805012205/NEC-V-ISCSI : FAILURE in 6h 09m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/917292/1/20240805012205/NEC-V-FC : FAILURE in 7h 13m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9c9d99af8377eded87568550692abbc568b19f3","revisions":{"d9c9d99af8377eded87568550692abbc568b19f3":{"kind":"REWORK","_number":1,"created":"2024-04-29 08:21:53.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/92/917292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/917292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/917292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/917292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/917292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/917292/1"}}},"commit":{"parents":[{"commit":"79629c346e001e88a03a63f3a713b7af6ee8193c","subject":"Merge \"Mark nexenta_encryption option as deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79629c346e001e88a03a63f3a713b7af6ee8193c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-04-28 18:48:58.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-04-28 18:48:58.000000000","tz":120},"subject":"wip: Add type field to backup API","message":"wip: Add type field to backup API\n\nChange-Id: I0276f4fe42478670aa24f3e725adbbcab086082d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d9c9d99af8377eded87568550692abbc568b19f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d9c9d99af8377eded87568550692abbc568b19f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
