)]}'
{"id":"openstack%2Fcinder~917651","triplet_id":"openstack%2Fcinder~master~I8f6b3393f8595e28a2c7e7d070f3b2434cf8a5e0","project":"openstack/cinder","branch":"master","topic":"reno-eom-zed","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-19 16:50:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f6b3393f8595e28a2c7e7d070f3b2434cf8a5e0","subject":"reno: Update master for unmaintained/zed","status":"MERGED","created":"2024-04-30 10:59:13.000000000","updated":"2024-07-19 20:19:38.000000000","submitted":"2024-07-19 16:50:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917651-reno-eom-zed","meta_rev_id":"782dff1256403e4ed16e8e86568468086ef138aa","_number":917651,"virtual_id_number":917651,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-19 16:50:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2024-07-19 14:51:25.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2024-07-19 14:51:25.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 11:25:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 11:32:08.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-30 11:36:55.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-30 12:41:23.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-04-30 13:09:59.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-04-30 15:06:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-30 15:09:54.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-01 17:37:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-02 20:54:05.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-11 22:36:50.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 13:55:32.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 14:14:02.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-19 14:51:25.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"728bc8fa08d17a9305a59d782831e1ec05dda7ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-04-30 10:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b518cda02a119fdc5ed10a2f156bb87e86ee298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:25:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34fc7b3c37d348248363029b3d58e9f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/139c82fb131c49cbba4ebdb35cad91f1 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d4e8255aba484e826889a23dadc679 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0016ff9606f541309ccf1c8e2d40a824 : SUCCESS in 17m 16s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b5dff300c9064420b49d09fc5141d8ab : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"007a1f1890d7a5ddc879f7f67437b8c917cd860a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-30 11:32:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_04_2024_11_26_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"027b9a87584d104e2ee76caeb916040fb25be780","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-30 11:36:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/1/upstream-check/cinder-tempest-plugin-ontap-fcp/cfe704d/ : RETRY_LIMIT in 4m 27s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/a058051/ : RETRY_LIMIT in 4m 26s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/1/upstream-check/cinder-tempest-plugin-ontap-nfs/521cbc6/ : RETRY_LIMIT in 4m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/795533f/ : RETRY_LIMIT in 4m 40s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1bc8c57e9a6369c70692ec20128bf4b2f71780c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-04-30 11:40:07.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9671ce9e66050f19316731f69778e1b95e1dd222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:58:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cee3abfdb404822aebc9bf35ee43e20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33813ceddf5945239c2a32146692c0eb : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e0f31bc4cd443bbb420e996fc3f411 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/723612f06a534b66ad42ba7e1ce29641 : SUCCESS in 16m 12s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/42dfca71ac4a4ee1b2133e68f216c1a4 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a1ea80eb2390dc2060abf3b901e908fa18c6a00c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-30 12:04:34.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_04_2024_11_58_28\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"7b968c1b61a53aec011c6b50504f02551e157873","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-30 12:23:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/2/upstream-check/cinder-tempest-plugin-ontap-fcp/08b1b9b/ : RETRY_LIMIT in 3m 47s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/ccb1e23/ : RETRY_LIMIT in 3m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/2/upstream-check/cinder-tempest-plugin-ontap-nfs/0859fc5/ : RETRY_LIMIT in 3m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/753a010/ : RETRY_LIMIT in 3m 35s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"268d4ae8733fd3162ed6cc361cbd4d4becaf63a8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-30 12:41:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/917651 : FAILURE in 23m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0fb6ac960184cac85e6371a2e5ef9f2f358492cc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-04-30 13:09:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/917651 : FAILURE in 24m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0152608b2c49ae06b186179181e3dd9db50dd017","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-30 13:09:59.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7bc641a5d3604d96b8fdef9b55381c86\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e9dfec4114a04635a320d8ec0c181e44 : FAILURE in 1h 27m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9e97b7e2f560d9c5e2d73e5432f8ba4fc7adb49b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-30 15:06:34.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_Unity_FC/244bc58/DellEMC_Unity_FC/1798 : SUCCESS in 1h 26m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_Unity_iSCSI/bbfff56/DellEMC_Unity_iSCSI/1826 : SUCCESS in 1h 37m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"207e26843624dc2f640317fa5d037870cce37f84","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-30 15:09:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_PowerStore_iSCSI/fdde74b/DellEMC_PowerStore_iSCSI/1290 : SUCCESS in 1h 17m 55s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_PowerStore_FC/4a79e36/DellEMC_PowerStore_FC/1461 : SUCCESS in 1h 02m 44s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_PowerStore_NFS/bd5842b/DellEMC_PowerStore_NFS/980 : UNSTABLE in 1h 11m 14s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_PowerStore_NVMeTCP/228498b/DellEMC_PowerStore_NVMeTCP/841 : SUCCESS in 1h 36m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"daca9f239ba8e38f081ab1f28032fa96b847fdfc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-01 17:37:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/917651/2/20240501215249/NEC-ISCSI : SUCCESS in 4h 20m 26s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/917651/2/20240501215249/NEC-FC : SUCCESS in 4h 44m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0b4fad45aa65abc4b06a3645b3752b62ce8e18a5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-02 20:54:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/324/.","accounts_in_message":[],"_revision_number":2},{"id":"3aee1fe654b77cc4da23f8c4b87e4558cb65f2de","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-11 22:36:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_PowerFlex_3.6/82aaf50/DellEMC_PowerFlex_3.6/1290 : UNSTABLE in 1h 42m 50s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_PowerFlex_4.5/cd17946/DellEMC_PowerFlex_4.5/219 : SUCCESS in 2h 02m 01s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/917651/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/b8db2d4/DellEMC_PowerFlex_4.x_NVMeTCP/164 : FAILURE in 2h 52m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"dfa6388a8abd8e0ccb1e5c2d705d006fea84252f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-19 12:40:20.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a223d31b9be7eeccf8791ce62c9d82f8a6b08980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 13:01:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4532de7e7a34cf69886b58fa684809f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a475e655bf942418318d38f44d0876d : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/851953d73f7547d0946335216abe7642 : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5ebae371b6d41a887235032b8419d09 : SUCCESS in 19m 28s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5a3f32a9bf164a868c03e003e2c7c7ed : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":3},{"id":"9787b7a71bc4996b1a49875f44a845a34ba0a37d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-19 13:07:34.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_07_2024_13_01_52\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"ba18dfd05bf5d7aff320cfb478c2293d7ec39dae","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-19 13:13:58.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/41e1da68fe7f46888bd8d7c96e363a0e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/cb61a626465846cda9da852187ce4fec : POST_FAILURE in 20m 54s","accounts_in_message":[],"_revision_number":3},{"id":"306d849b6a554ae3cff6aca895ab38f9131cc623","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-19 13:14:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/829f71c/ : RETRY_LIMIT in 3m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/917651/3/upstream-check/cinder-tempest-plugin-ontap-nfs/ad57031/ : RETRY_LIMIT in 7m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b5e20b7c3be45b9487fa332335e3f3030357593d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-19 13:55:20.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/917651 : FAILURE in 22m 11s","accounts_in_message":[],"_revision_number":3},{"id":"48a86164851bef6012f4bdd8b97a41f6177fa99a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-19 13:55:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/917651 : FAILURE in 15m 04s","accounts_in_message":[],"_revision_number":3},{"id":"18a4154c7c27f672a6f35ac5f2afce957629f43a","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-19 14:14:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/917651 : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":3},{"id":"30c5518cc0d4ddd7404e1bbf96451d7dab786ff6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-19 14:43:40.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_PowerStore_iSCSI/5096e37/DellEMC_PowerStore_iSCSI/1526 : SUCCESS in 1h 14m 45s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_PowerStore_FC/5152726/DellEMC_PowerStore_FC/1723 : SUCCESS in 58m 45s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_PowerStore_NFS/ec046d6/DellEMC_PowerStore_NFS/1221 : FAILURE in 1h 16m 49s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_PowerStore_NVMeTCP/7694601/DellEMC_PowerStore_NVMeTCP/1080 : SUCCESS in 1h 41m 43s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_PowerStore_NVMeFC/e627df3/DellEMC_PowerStore_NVMeFC/42 : FAILURE in 8m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"05d162cb9d8df2a783e8755c0d900e587898a4e9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-19 14:51:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f4bd8281adf07426ba9cd9042a3afa23aff8e214","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-19 14:51:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/759/.","accounts_in_message":[],"_revision_number":3},{"id":"b3e543a95d8cf836ca61d42cac4bf4c374dc00ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:52:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f011577efe4eb306533e9454ef95eddc86a529c5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-19 15:16:24.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/275/.","accounts_in_message":[],"_revision_number":3},{"id":"2d0dda8001075261decb149c0d2079c581652f67","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-19 15:16:34.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_Unity_FC/6dadc33/DellEMC_Unity_FC/2263 : SUCCESS in 1h 33m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_Unity_iSCSI/ede7227/DellEMC_Unity_iSCSI/2270 : SUCCESS in 1h 47m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2d20c828b68f00b4b79a2244de878b5907ca2e03","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-19 15:33:40.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_PowerFlex_3.6/ca7bce0/DellEMC_PowerFlex_3.6/1751 : FAILURE in 2h 04m 18s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_PowerFlex_4.5/54e3631/DellEMC_PowerFlex_4.5/630 : SUCCESS in 1h 39m 47s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/51/917651/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/42a7602/DellEMC_PowerFlex_4.x_NVMeTCP/588 : FAILURE in 13m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"40286acb45dc1e2b7e1c90814a9660722033aeb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:50:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e14a36ae51d742bc8124c108658a2106\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c26230b0e10418584dad76d0d9764af : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70313a0e72e64b11828a3058b34c3ef4 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f18443fe3ab24f45881c567b4cb0804b : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":3},{"id":"fabcb762a95cca68df18a8b08e2184792cfe69e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:50:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cd94f62bb1b9b038ffec3872a14dae228d46b1ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:51:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797682d7bb3c4448b6372263b1186063\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb34318dbb54f218a62c63f7452f464 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aa003bd70f2455984f568572a93512d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3},{"id":"782dff1256403e4ed16e8e86568468086ef138aa","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-19 20:19:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/917651/3/20240719233420/NEC-ISCSI : SUCCESS in 5h 21m 46s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/917651/3/20240719233420/NEC-FC : SUCCESS in 5h 45m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f7b0950deabb7ee49d0271bf02c6c65f8d67757c","revisions":{"6604db8286708aaa404aebc6df7e699529df0c7c":{"kind":"REWORK","_number":1,"created":"2024-04-30 10:59:13.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/51/917651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/917651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/917651/1"}}},"commit":{"parents":[{"commit":"79629c346e001e88a03a63f3a713b7af6ee8193c","subject":"Merge \"Mark nexenta_encryption option as deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79629c346e001e88a03a63f3a713b7af6ee8193c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-04-30 10:59:11.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-04-30 10:59:11.000000000","tz":0},"subject":"reno: Update master for unmaintained/zed","message":"reno: Update master for unmaintained/zed\n\nUpdate the zed release notes configuration to build from\nunmaintained/zed.\n\nChange-Id: I8f6b3393f8595e28a2c7e7d070f3b2434cf8a5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6604db8286708aaa404aebc6df7e699529df0c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6604db8286708aaa404aebc6df7e699529df0c7c"}]},"branch":"refs/heads/master"},"eb461b28dc11b029de37cfb0b54c653d62a96e6b":{"kind":"REWORK","_number":2,"created":"2024-04-30 11:40:07.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/51/917651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/917651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/917651/2"}}},"commit":{"parents":[{"commit":"79629c346e001e88a03a63f3a713b7af6ee8193c","subject":"Merge \"Mark nexenta_encryption option as deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79629c346e001e88a03a63f3a713b7af6ee8193c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-04-30 10:59:11.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-04-30 11:40:03.000000000","tz":0},"subject":"reno: Update master for unmaintained/zed","message":"reno: Update master for unmaintained/zed\n\nUpdate the zed release notes configuration to build from\nzed-eom, which is the last point at which the cinder project\nteam had responsitibility for the branch.\n\nChange-Id: I8f6b3393f8595e28a2c7e7d070f3b2434cf8a5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb461b28dc11b029de37cfb0b54c653d62a96e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb461b28dc11b029de37cfb0b54c653d62a96e6b"}]},"branch":"refs/heads/master"},"f7b0950deabb7ee49d0271bf02c6c65f8d67757c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-19 12:40:20.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/51/917651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/917651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/917651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/917651/3"}}},"commit":{"parents":[{"commit":"79629c346e001e88a03a63f3a713b7af6ee8193c","subject":"Merge \"Mark nexenta_encryption option as deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79629c346e001e88a03a63f3a713b7af6ee8193c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-04-30 10:59:11.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-07-19 12:40:20.000000000","tz":0},"subject":"reno: Update master for unmaintained/zed","message":"reno: Update master for unmaintained/zed\n\nUpdate the zed release notes configuration to build from\nzed-eom, which is the last point at which the cinder project\nteam had responsibility for the branch.\n\nChange-Id: I8f6b3393f8595e28a2c7e7d070f3b2434cf8a5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7b0950deabb7ee49d0271bf02c6c65f8d67757c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7b0950deabb7ee49d0271bf02c6c65f8d67757c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
