)]}'
{"id":"openstack%2Fcinder~918316","triplet_id":"openstack%2Fcinder~master~I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66","project":"openstack/cinder","branch":"master","topic":"user-visible-metadata-in-volume-types","attention_set":{"34420":{"account":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"last_update":"2025-01-13 12:46:30.000000000","reason":"Some votes were not copied to the current patch set"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-01-06 12:04:01.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"28271":{"account":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"last_update":"2025-01-06 15:00:08.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66","subject":"Add volume type metadata table to database","status":"NEW","created":"2024-05-06 13:14:47.000000000","updated":"2025-01-13 12:46:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00788cadd633d24c215070bc126e36416c2d113d","_number":918316,"virtual_id_number":918316,"owner":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-06 14:47:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37269,"deleted":true}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-06 13:33:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-06 13:44:10.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-06 14:39:39.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-06 14:54:17.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-06 14:56:13.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-06 15:28:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 04:38:15.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-07 09:48:35.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-05-07 10:13:24.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-07 11:46:34.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-08 10:41:33.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-05-12 20:50:15.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-05 07:26:45.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-05 07:26:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-05 08:43:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-05 12:28:06.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-07-19 09:01:33.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 13:08:23.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-05 09:00:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-10-08 11:51:51.000000000","updated_by":{"_account_id":37269,"deleted":true},"reviewer":{"_account_id":37269,"deleted":true},"state":"CC"},{"updated":"2024-10-21 08:00:10.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-11-06 16:30:27.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-11-12 13:36:14.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"CC"},{"updated":"2024-12-18 08:04:36.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2025-01-06 13:08:19.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"e0777b43f1124b97e9d61fb0cb9ddbf091f6cbef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-05-06 13:14:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1bce0f2c5622e813da089615c9a3bca9c9674fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 13:33:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ec59a333d047e4af657bb8e9d30481\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fabc59e1e33f4b48b1de79be6d56fcb8 : FAILURE in 9m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04ff02ec568b4c3f99cc6657c203e3c0 : FAILURE in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9170b36f41fb0a64871dea049484a72f6aae6f31","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-06 13:44:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 27m 45s","accounts_in_message":[],"_revision_number":1},{"id":"34d397d98a79fc4a811c0d65595e36149caea36f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-06 14:13:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 25m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bdb55289aa6efa6d8357f9d14ea5367dfa414e2c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-06 14:39:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/335/.","accounts_in_message":[],"_revision_number":1},{"id":"d34b8d1fcc282b9c780a2bebd330bee19d2786b7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-06 14:54:17.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3d19838dc6294b4c95f144f11f201c77\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0a85f82a0db3431f8955fa004fa4f606 : FAILURE in 1h 31m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cf935aa87463cd623e9db52278b6f5a063684d2c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-06 14:56:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/1/check/DellEMC_Unity_FC/099c427/DellEMC_Unity_FC/1810 : SUCCESS in 1h 20m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/1/check/DellEMC_Unity_iSCSI/5dd9f34/DellEMC_Unity_iSCSI/1838 : SUCCESS in 1h 41m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7e3f5be671cd35543b2ca26ac60fafe98591f1fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 15:28:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bb7740527c34638a77c2c7d734c2fc5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/467d5e090d5341438f5a9e52f3ad9baf : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6610894f40534ee599141cadc78ce8e8 : FAILURE in 12m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c3f59555ff74452b8319633ff8523ca : FAILURE in 12m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94f2e0d9473f450c986d7ed8c0ee26dc : FAILURE in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c63701beb48441793feb9a47d6f3721 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/7650dbadd4f14d70b4129d11c7aa902c : SUCCESS in 1h 00m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18060b0af90f4663b000131477b56a80 : FAILURE in 1h 21m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55cd322dd21644d7a9cc3f3cd2aa1fe0 : SUCCESS in 1h 10m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b397bdb815724effbef36b0ff44d1971 : FAILURE in 12m 00s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e4e814f1f0d549c0ba00da8762a10134 : FAILURE in 10m 58s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/828f3d5e5d094589be0c8cbd10065180 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/140fda505aa2417bb4348e477fcd8394 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc04bc181e4c4f27abc9545b3709b063 : SUCCESS in 5m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d80f06f394db4604bc2046c12c55b4d7 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c3bf302a88d14b4da6205f704df61c3a : SUCCESS in 35m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78a68be617c649d7816ac42b1ff773d5 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c43bb8d616534e108733123064701408 : SUCCESS in 1h 28m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9864cb7a0e914cb8b85f435f2caaa45f : FAILURE in 1h 41m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d56ac19877c74e1786d49da3ff4d6c10 : SUCCESS in 2h 07m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/00f561e14d0e4c558ea5911c8400b1df : SUCCESS in 1h 54m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ac548409df6a4572a86467371128b311 : SUCCESS in 40m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3f0443b327824633a1915fcb07c793a9 : SUCCESS in 1h 18m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/20c7be0c558840dba15f67306cf9af28 : SUCCESS in 1h 25m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d7ba56c955f440d598b015f54f6ae7fb : SUCCESS in 53m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/809dd3027cb04bdf9e096b297410854e : SUCCESS in 1h 23m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e6046726e4624aa8827e6dfc4cc60ece : SUCCESS in 1h 29m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9843a10c3de4c2eaae5c0167d9584a9 : SUCCESS in 1h 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05fdd3c2fa3d4cfab05fc1a0b3a284d0 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b358e6d47774840750e0878b1a93a3655fd89fe2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-07 04:38:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/1/20240507085120/NEC-ISCSI : SUCCESS in 4h 15m 20s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/1/20240507085120/NEC-FC : SUCCESS in 4h 46m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7f8c9ff3d9db5435954f8c3e9c12bf29a338a5bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-05-07 07:27:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1c6f72b83a517480ffcdfcba1d6e5edb258c18ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 07:37:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac60722d2e8471286111dcc1371313e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af5c2e9cd48d4b199dba4ff13a4d3f52 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cf28cbc32939405c863efe97517b0536 : SUCCESS in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e2542e19207a6ff5e8a3d9438d61c71478ebb0b8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-07 07:52:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 23m 22s","accounts_in_message":[],"_revision_number":2},{"id":"329313661df0600bd6bbba14af3b6adc2a594235","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-07 08:21:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 24m 14s","accounts_in_message":[],"_revision_number":2},{"id":"36e3f4f6760f0f312f2d2c5a7c47ea4801576f77","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-07 08:25:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/28/.","accounts_in_message":[],"_revision_number":2},{"id":"30eb3bb43acb771ad7695713cace5181399919a1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-07 08:51:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/339/.","accounts_in_message":[],"_revision_number":2},{"id":"0c8a8d45ff2555c72338b02c1aec0d3cb7c93533","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-07 08:55:53.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/37fa9869b8354f25980dad3a00720d1f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2398e2a8a08042759762b0298da38221 : FAILURE in 1h 25m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a2acb807e0a6884d1ac8eba906d15f99e9cf72b2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-07 09:06:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_Unity_FC/0175630/DellEMC_Unity_FC/1816 : SUCCESS in 1h 24m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_Unity_iSCSI/55b9615/DellEMC_Unity_iSCSI/1844 : SUCCESS in 1h 38m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a52c96d35eedae90149b80d49c9705166b9af49c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 09:42:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/002ec0d970fe492bac573e2402499080\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb68c4c49da40f892fe03618d03c1f8 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6a2b23dd45e44a6a83b46252490ef3a : SUCCESS in 11m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa78ca059eb94b19bc8b63c9fae4d5b7 : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5e1407e726a4e399d4f6050ca9d8317 : FAILURE in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ef7181b5b54d93b43b2d781fcc314a : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/5fde00ac841c40eaaf7b056c38321654 : SUCCESS in 1h 07m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c5afc1015bf4a87839657e11cb673eb : SUCCESS in 1h 37m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1e595798c4d4bd6be62f2e23ea6e581 : SUCCESS in 55m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b3222a09107f46c192f1e709fb327a26 : SUCCESS in 14m 47s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b0154917f2f34703ab3157931fc8557c : SUCCESS in 12m 39s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/845959dceb79452ca0fa9074814efc4e : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b714d750f4342edbb8c3bbd0b1a2f28 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b1009404eb674db7bf3ff80012a7603d : SUCCESS in 6m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/153ae286b1e8418f9f380cc2d611180a : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f093898339b0467db6d4af271435405d : SUCCESS in 35m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec4054095780407d986d9365ffc80024 : SUCCESS in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/69342f09fd3546f983ef3dfe69e04bb2 : SUCCESS in 1h 31m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fea2128ffd104cc28c894b8b46ff9879 : SUCCESS in 2h 03m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c06aee279d0f4f5fbbbe14e2449b28b3 : SUCCESS in 1h 33m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/900d3e59c0144d9985e9bac16b6043bf : SUCCESS in 2h 01m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/feb3ed3fffd94870875c521a8d298a6a : SUCCESS in 39m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c2634750a02429fbbd4fde07af98812 : SUCCESS in 1h 15m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1638bc36aa8e4d0d98cd522335de7f2d : SUCCESS in 1h 19m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bcce98b86b2c46329bfd1e76e1797067 : SUCCESS in 37m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b298bf0691834b88b9a3159ee2a2fa11 : SUCCESS in 1h 08m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/188cdf220f344e4c98bd56420c851371 : SUCCESS in 1h 20m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eca7ba9abfbc4e4694361e2164fc8768 : SUCCESS in 1h 39m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd4446e0821d4d44ab2bc8aeedc1d0b7 : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":2},{"id":"97f0eb4c5429d68236b44691a87be4edf92ed45b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-05-07 09:48:35.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_05_2024_09_42_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"3f2543c9f2fefffa8a4770a8135a0b5dc7d99a6d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-07 10:13:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/918316/2/upstream-check/cinder-tempest-plugin-ontap-fcp/69f3f08/ : RETRY_LIMIT in 3m 36s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/918316/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/c290789/ : RETRY_LIMIT in 3m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/918316/2/upstream-check/cinder-tempest-plugin-ontap-nfs/c75c39a/ : RETRY_LIMIT in 3m 50s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/918316/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c9e8598/ : RETRY_LIMIT in 3m 32s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e45c83aa18e7334a346464b4c0fe711bac27bdf6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-07 11:46:34.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_PowerStore_iSCSI/c14d349/DellEMC_PowerStore_iSCSI/1300 : SUCCESS in 1h 18m 12s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_PowerStore_FC/b74cf1c/DellEMC_PowerStore_FC/1470 : SUCCESS in 1h 02m 48s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_PowerStore_NFS/aeacd3a/DellEMC_PowerStore_NFS/990 : SUCCESS in 1h 14m 59s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_PowerStore_NVMeTCP/85cec4c/DellEMC_PowerStore_NVMeTCP/850 : SUCCESS in 2h 04m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"dbb02b24b5fa892780bd01bc92678b58ba322740","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-07 12:18:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_Unity_FC/1f9c445/DellEMC_Unity_FC/1818 : SUCCESS in 1h 16m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_Unity_iSCSI/80b3455/DellEMC_Unity_iSCSI/1846 : SUCCESS in 1h 33m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7b862aa94b8e5b76a08e98ad5f7a681e078cb29c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-08 04:30:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/2/20240508040829/NEC-ISCSI : SUCCESS in 9h 12m 21s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/2/20240508040829/NEC-FC : SUCCESS in 9h 21m 39s","accounts_in_message":[],"_revision_number":2},{"id":"205e27ead7d1cb2c9b16e336b94a73a30e7c6d7a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-08 10:41:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/2/20240508161231/NEC-V-ISCSI : SUCCESS in 3h 13m 25s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/2/20240508161231/NEC-V-FC : SUCCESS in 3h 28m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3777110dc208602d757aed5fc011f8ade2fbc05a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-12 20:50:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_PowerFlex_3.6/a8e96db/DellEMC_PowerFlex_3.6/1303 : SUCCESS in 1h 39m 08s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_PowerFlex_4.5/c9edf82/DellEMC_PowerFlex_4.5/229 : SUCCESS in 2h 09m 15s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/9d6b6f3/DellEMC_PowerFlex_4.x_NVMeTCP/175 : FAILURE in 2h 54m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"5a3631f1a99455ded1e72b6f28cfe32cbdca4f8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-05 07:26:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2082ab8f8ad2429d91e4f27276f5062f2b00c91c","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-05 07:26:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"35676a0fd53809f9e79df15789a49ea156cc975f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-05 07:26:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"5ccfb1c04a404c336d740431634d518e7c1efd04","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-05 07:26:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"d57bc4c35d991d2375f425bff073c54789f0361b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 07:35:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5af44515ef1c436abb56dec8a4aaa8bc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98c03a040fdf489286149c163c61996e : FAILURE in 8m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47b759ba9a2b45609b0acd9bf3e427d4 : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"771401c39ac86e6009a41d81856ac7f8774e9d5a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-05 07:51:55.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/991739e55fa04579aa3ada75c6c9332a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/844bde55c08f4abeb3245ccb1e8dbbb6 : POST_FAILURE in 22m 40s","accounts_in_message":[],"_revision_number":3},{"id":"cb9166c7644eaec5694e95b728c8e0754bafed21","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-05 08:43:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/226c793603144a378d0640090c16b8f8 : FAILURE in 1h 14m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5bf7bf2bef554d38ae5a847243c84713 : FAILURE in 1h 15m 46s","accounts_in_message":[],"_revision_number":3},{"id":"352ac3cd279de45205cfe8a52ed82de4cb14cb5f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-05 08:54:58.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/682/.","accounts_in_message":[],"_revision_number":3},{"id":"ab7fe72873e01fc4b2f31d7cac903d0dc6bdee64","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-05 09:16:33.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/3/check/DellEMC_Unity_FC/50f4c86/DellEMC_Unity_FC/2178 : FAILURE in 1h 50m 05s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/3/check/DellEMC_Unity_iSCSI/692ec6a/DellEMC_Unity_iSCSI/2187 : FAILURE in 1h 48m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"273f223e70b3792da221f26a26ab5deb8cf85b1c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-05 09:16:44.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/3/check/DellEMC_PowerFlex_3.6/4dcf44b/DellEMC_PowerFlex_3.6/1664 : SUCCESS in 1h 47m 16s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/16/918316/3/check/DellEMC_PowerFlex_4.5/8439ede/DellEMC_PowerFlex_4.5/564 : SUCCESS in 1h 50m 18s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/8f7f45b/DellEMC_PowerFlex_4.x_NVMeTCP/516 : FAILURE in 11m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"b7861ca764441a1351fa5270b1f6f2979c5020d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-05 09:23:43.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/185/.","accounts_in_message":[],"_revision_number":3},{"id":"9b7e1cb40255f26b41e1c6bc37913bee711e039d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 09:46:57.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31d6e40c704643a6b0a6a1aae61bfe79\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b69612c33e424293d5f937eb30cca3 : FAILURE in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef1301bb5b9b408ab94b7a22ddeae1c9 : FAILURE in 9m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ed672edf855419bb188cc1d5893905d : FAILURE in 10m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52225327a40a4f92b52b6896f3a03c52 : TIMED_OUT in 40m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54965c2cc0884fa7adcbe2ce58a122f2 : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/df6fe37e26bd4f9cb5e207beb8621183 : SUCCESS in 55m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/419a70bda9484e5bb2fc8570e7029267 : FAILURE in 1h 22m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be223dffd3db419881066c295ac93a93 : SUCCESS in 59m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8330ac698c36448ab43d5a9c1a51e769 : FAILURE in 13m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/049d3671c5c24881ae192a294f8ba0c4 : SUCCESS in 4m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d338047ecc342a0a23d68e4d715b4a4 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/41eeb27034994e94924c08098e15ee52 : FAILURE in 5m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/707705819e7248cc93d5d39d16c94d6f : FAILURE in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b7189dabc5f64f82b2fa057a2189ee06 : FAILURE in 42m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38c7a9fba86746dbb82da7fa25bded95 : SUCCESS in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/439ce3597abc475bbee81280ca776fc4 : FAILURE in 2h 05m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/688d32bbfae8428b9294d8eb02a4ebea : FAILURE in 2h 20m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/029f35a9b4434dd2b721507d00fe35c8 : FAILURE in 1h 55m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/edf862de46de4147a4dd9d389573d443 : FAILURE in 1h 57m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e2b6da8a20814718a9650bfb8a14c7e1 : FAILURE in 32m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/21dcf01b1f7545efbe472837bf24988a : SUCCESS in 1h 11m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09eaf318b55b426d9a0955587e274ec0 : FAILURE in 1h 46m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c2430bfbfced44e2be6ff564ef885888 : SUCCESS in 52m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/03c7364828c148ffb026f2fcf7fd90e5 : SUCCESS in 1h 32m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/beadd345bc5747b99ac9ecbca291a154 : SUCCESS in 1h 30m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a55ea52a5ec451fa359d86c5a0ced58 : SUCCESS in 1h 37m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd07edefd4c24d00a295e2dc53ff6be5 : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":3},{"id":"91c02ca5f622ab9bd9ac28396f74e7df61611ec7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-05 12:28:06.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-3/FC/ : FAILURE in 5h 1m 44s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-3/iSCSI/ : FAILURE in 4h 33m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"aaba28c3275de23da5594b8c68b9753e25f17645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-05 12:55:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"de85c033fe8f8609b07a540a0ae92ca3b3511da3","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-05 12:56:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"7101607c3baf8f39ce80be8a537b08433108870e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-05 12:56:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"d7c924758bd9a3b398c38c60d506e8f0ba67c5ff","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-05 12:56:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"a69c7cfaaba1f8c94cdde53b9059bf0063401842","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 13:02:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c8d4ff971c44e99e5e46d8a048546f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1127c536040740c49eedab5cd070d0b1 : FAILURE in 3m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b75745dbc8994be0a5a1c74e2f1bb8e6 : FAILURE in 3m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"adc76e3479a16ccf8e5718f33d923a9a820c611e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-05 13:19:27.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/73ecdd76faed44eeab77d2fc3a98b1e1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2fa576301993405e91aefbe6cab6a6df : POST_FAILURE in 20m 24s","accounts_in_message":[],"_revision_number":4},{"id":"27ffccd313293c9fa2fd2702d7478b810f65a9e4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-05 14:17:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f682e6a852154d57ad7ffe6d05553539 : FAILURE in 1h 19m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/32b8ee36a9e1415d9f08c08c0a80b7d3 : FAILURE in 1h 17m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d5f50a584cce4be83050fb717b3a2fd8d448b9f4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-05 14:28:45.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/685/.","accounts_in_message":[],"_revision_number":4},{"id":"c7188330ad6a206d4f218b5ff14864ccb1941eef","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-05 14:45:08.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/189/.","accounts_in_message":[],"_revision_number":4},{"id":"cbe8c6e747cf5f5ee938ec13c2e45c7abd64a6df","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-05 14:48:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/4/check/DellEMC_Unity_FC/e581483/DellEMC_Unity_FC/2180 : FAILURE in 1h 24m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/4/check/DellEMC_Unity_iSCSI/24c0b23/DellEMC_Unity_iSCSI/2189 : FAILURE in 1h 52m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"59339fd21d76b5e78cffe3bb66d437a572eff33b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-05 15:06:56.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/4/check/DellEMC_PowerFlex_3.6/73cc858/DellEMC_PowerFlex_3.6/1666 : SUCCESS in 2h 10m 49s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/16/918316/4/check/DellEMC_PowerFlex_4.5/5321bf7/DellEMC_PowerFlex_4.5/566 : SUCCESS in 1h 59m 11s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/47ff2a8/DellEMC_PowerFlex_4.x_NVMeTCP/518 : FAILURE in 11m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"a5d36925f6dd9a90b51467c928c6ef7c0c20ba5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 15:28:29.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e1745aeed75497f89f2746f506842a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f3c8dab77b49aabdf8b2793db428b3 : FAILURE in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1dedbebb2424a1084c4507315f18c71 : FAILURE in 11m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/762c4b4f386a4a6f8a15cff11d76c8b9 : FAILURE in 9m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e765faccae0474f8d66dae5d0220618 : FAILURE in 15m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12c67b166104f58939e9fc59da43e5a : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/baf10bc83eb24ddba309592798d967de : SUCCESS in 49m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22b46d073c5d432db298fd739bdca565 : FAILURE in 1h 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22e0bc1e2d8e4a3da06aead1f0969ce3 : SUCCESS in 53m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fb8da5d9a24a45eeafa4aaf7e265f065 : FAILURE in 13m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5abae355e9c644989969860edc710d1e : SUCCESS in 7m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e5bfce45ac6480e82d676ba59888fe5 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/86314e032d9a4a8881f84c5155c902c1 : FAILURE in 7m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1e1d3cff73fa4444a1155ce0a3ec2365 : FAILURE in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d56957d81cf64034b6d5e65b3da9b777 : FAILURE in 44m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/348529acc79b434cbef7c2d300b7a5de : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7578b99f66484af7b4db882b2a787c83 : FAILURE in 2h 02m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a28fd808da0c42b4a9ad87d51324329d : FAILURE in 2h 31m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ab6391627974ba49a0de68eb1afa6bc : FAILURE in 2h 16m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/33db414f92fd4c429928d4a51db66291 : FAILURE in 2h 01m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0cfae86d7ec14262bd9de1b4ff3259fa : FAILURE in 42m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7999c40f815945de978ca14119b99bcf : SUCCESS in 1h 13m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a2b562dbd1fe425386dec6a7b2675b95 : FAILURE in 1h 53m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8dad083fb0734b579c09c70b55313609 : SUCCESS in 35m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f3ac1e1826f745358d2f8997f9a3f69c : SUCCESS in 1h 29m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4c14832e88c64d54847e6caafab21418 : SUCCESS in 1h 35m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68c0251929484c0f8a6c3bbe87c6df23 : SUCCESS in 1h 39m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9498e3ed76e5488abf2e5b57c4d275bf : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":4},{"id":"c82003e40bd8ee23bcca68d582e6d8c5c3be9dca","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-05 18:11:54.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-4/FC/ : FAILURE in 5h 15m 47s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-4/iSCSI/ : FAILURE in 4h 19m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"0a4add36ddbcd2d560d6b071a282b30766e1e348","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-06 11:58:34.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/4/20240706113230/NEC-ISCSI : FAILURE in 9h 03m 49s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/4/20240706113230/NEC-FC : FAILURE in 9h 26m 02s","accounts_in_message":[],"_revision_number":4},{"id":"2011cac6ccce8deabcca2019081ead8bf326668c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-07 11:04:46.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/4/20240707035324/NEC-V-FC : FAILURE in 7h 32m 03s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/4/20240707035324/NEC-V-ISCSI : FAILURE in 16h 11m 19s","accounts_in_message":[],"_revision_number":4},{"id":"0b4eb105d50142a3575330eb29342a17de31c728","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-19 07:04:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"97a94f7b0e4a20bb1716c1a3057cd8f0f784fe97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 07:20:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a168093ce01432ab320c5b11c989b27\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce24ca4a15a14660ade0cd55df422129 : FAILURE in 10m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/61f1071ef4a94e3eaf3c5dc34e383ad3 : RETRY_LIMIT in 3m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9c7242efc65f716ac9a0d68bdbe38100d6988f74","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-19 07:32:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 26m 51s","accounts_in_message":[],"_revision_number":5},{"id":"4c7c49105bc9ca379b944965fa85f5f318e56716","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-19 07:37:45.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f6b4af211aef40b1a1c1778527b2891c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/98aab09a4daa42f5ab4f74c25a42aa1a : POST_FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":5},{"id":"59906bb46562caf675c63bc52a314f5d515fe769","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-19 07:42:28.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 36m 21s","accounts_in_message":[],"_revision_number":5},{"id":"5fc1b7ff105c691db2cec9af50c37dc6f2828e84","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-19 07:49:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 35m 52s","accounts_in_message":[],"_revision_number":5},{"id":"de6cbb2a03c2bddbf2125be1cea323533788a92b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-19 08:26:46.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/754/.","accounts_in_message":[],"_revision_number":5},{"id":"f7c34f23272ce6fbcc595c0b6136d29e9d604d3f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-19 08:29:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/375aa82e728d4253851545ad2e8e2157 : FAILURE in 1h 22m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/217149abe87b473abe2deb91ac918790 : FAILURE in 1h 21m 04s","accounts_in_message":[],"_revision_number":5},{"id":"899f823f43ec933d42304dfacd4ae6898a498093","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-19 09:01:14.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/5/check/DellEMC_PowerFlex_3.6/70d572d/DellEMC_PowerFlex_3.6/1746 : SUCCESS in 1h 45m 10s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/16/918316/5/check/DellEMC_PowerFlex_4.5/2a8dda3/DellEMC_PowerFlex_4.5/625 : FAILURE in 1h 56m 20s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/b22d755/DellEMC_PowerFlex_4.x_NVMeTCP/582 : FAILURE in 14m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"37d61cfd570b15e0d157e848ca2040ed7a4e7065","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-19 09:01:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/16/918316/5/check/fujitsu-eternus-dx-iscsi/26bd4cc : FAILURE in 1h 56m 18s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/16/918316/5/check/fujitsu-eternus-dx-fc/257bf18 : FAILURE in 1h 52m 39s","accounts_in_message":[],"_revision_number":5},{"id":"dd08e865b0257456ac25b6cecb46b920f752d4b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-19 09:01:51.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/5/check/DellEMC_Unity_FC/b7636d3/DellEMC_Unity_FC/2258 : FAILURE in 1h 34m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/5/check/DellEMC_Unity_iSCSI/2c7fcf0/DellEMC_Unity_iSCSI/2265 : FAILURE in 1h 56m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"c8ab888192834361c9e427ad0e777b95f7bd0b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 09:34:05.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3a0a6e5dacc42f4b84c39d2380e3e77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7addd34766ad43809248669640305c55 : FAILURE in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f935964b5d142a4835b58dc3faa6a6a : FAILURE in 10m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44fefebd1a9f494f97ec9b80cdd3fd2a : FAILURE in 10m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea48799115ba4513bda4d22584277382 : FAILURE in 14m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d5b7eba24b4af087b9c5947a7bd605 : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/a70be9b953ec4249b898add022086d58 : SUCCESS in 59m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de7c62db2a924999bb16ba1a5ee462b8 : FAILURE in 1h 16m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/697e4debd2fe48dcaae0f4a032d48efb : SUCCESS in 57m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4cddaccc188044aa8372a54d979752e4 : FAILURE in 13m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ec47835637dd43dfa5cdb8ed33079129 : SUCCESS in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3404a76f9fd43d79398c7081af67741 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e8e6b598bd7e439d8086574997507f44 : FAILURE in 7m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/058cbc4be449450c83e83845597c653f : FAILURE in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/18d7b68c893c49aa830bf7004bd9ca26 : FAILURE in 37m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ad1c2736b6644cc9175c76fef811560 : SUCCESS in 6m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/10c3aa80b50245e895a64248737a2a2a : FAILURE in 1h 34m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b7ac3117175b4d708f54aadfc9276f64 : FAILURE in 2h 26m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/24550a0ddf774cc38d11eb357e3c21e4 : FAILURE in 1h 44m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/487e91e7804043aea7eadea77cc01e48 : FAILURE in 1h 52m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/219c1f4cc9314b02a3c22f29e8e5c06d : FAILURE in 24m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e941df4ff1d54b149a588b0c583e8e25 : SUCCESS in 1h 07m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/583baaa02bd04f578ec6e4a1540499a0 : FAILURE in 1h 29m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cd52262dbb4a486882038f1785c14711 : SUCCESS in 51m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/be29cbabbfc5452e8670e26fdbbe3a85 : SUCCESS in 1h 15m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2f4f2b42ad5f41668fa8917d069f7b57 : SUCCESS in 1h 20m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da9173c53d664165b7ff8ed79bf9ca21 : SUCCESS in 1h 32m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfc81aefce434878b58963022774e0ee : SUCCESS in 43m 18s","accounts_in_message":[],"_revision_number":5},{"id":"75c4536ca17bbb753c70d5fb0687b4e4f133ae46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-19 10:05:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4afc14454413fdca717ea9ebba9155ffc463223a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 10:18:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e302b261bb24a879b85ee98231d44d5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/facfce28a22a4ad6bd077880cf489f34 : FAILURE in 8m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e3b3c41d8952423ebd311ec315561b69 : FAILURE in 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3eae802c35d51c607d9b1b653e0020d51834c161","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-19 10:45:57.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 38m 41s","accounts_in_message":[],"_revision_number":6},{"id":"dc9779365c0754f5813b57ddd98d21e8d13834b0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-19 10:49:25.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 42m 12s","accounts_in_message":[],"_revision_number":6},{"id":"7aaf47da0c255b43dd405e9921ce5932d5747df1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-19 10:51:38.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 36m 52s","accounts_in_message":[],"_revision_number":6},{"id":"13a8601ffb94cf0592433f2c1c93d43bee35a3d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-19 11:00:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c74cfba25f3d48d30a8d54f7c55264f89a9cd47c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-19 11:06:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4756cf04f79bb05baed4be1635e5202d3974297d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 11:12:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7baaa643f68848ee8677d78f12d87640\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e60335a53924a749417a5dee9b8c402 : FAILURE in 4m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2641ebb5c3d24d15bd7fb7793199ecc1 : FAILURE in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"30f8c0989285f2c99d48b1b352fe1e38e7444b5c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-19 11:25:42.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/273/.","accounts_in_message":[],"_revision_number":8},{"id":"30b3dde520a4fb321f0b867b7de6bea5e13e941a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-19 11:31:01.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/8/check/DellEMC_Unity_FC/7d07b50/DellEMC_Unity_FC/2261 : FAILURE in 14m 49s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/8/check/DellEMC_Unity_iSCSI/0450c8e/DellEMC_Unity_iSCSI/2268 : FAILURE in 23m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"99a1adecd9a7e6561ba61796b787884c20f59120","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-19 11:31:59.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/610d3d6bd6ea436a8b87586d98fe4960\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1a39429d016a4483b928bcfaa459d995 : POST_FAILURE in 20m 36s","accounts_in_message":[],"_revision_number":8},{"id":"75863e88f768bc0ac5dc952a77b510f65f6690fa","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-19 11:43:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 35m 25s","accounts_in_message":[],"_revision_number":8},{"id":"762f36baf8945b768a31b52a4a6026024a9f0774","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-19 11:47:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f0a83d8b4afd4761922b1ca07e9ccfe0 : FAILURE in 39m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/422cac6a01884cdfa5cfa84efd60314a : FAILURE in 38m 04s","accounts_in_message":[],"_revision_number":8},{"id":"1f91e017ee80af230b875484340b17ea42096629","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-19 11:50:14.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 39m 12s","accounts_in_message":[],"_revision_number":8},{"id":"57dd8a9f4e44b3e3017c2de82a49ea55785191c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-19 11:58:33.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 38m 17s","accounts_in_message":[],"_revision_number":8},{"id":"17500393bdf138e117482aae34450890e810b1f0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-19 12:00:05.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/757/.","accounts_in_message":[],"_revision_number":8},{"id":"b42624723d0ad754adcab29549576d0d3ee7f877","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-19 12:00:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/16/918316/8/check/fujitsu-eternus-dx-iscsi/584ef46 : FAILURE in 51m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/16/918316/8/check/fujitsu-eternus-dx-fc/0d9569a : FAILURE in 52m 42s","accounts_in_message":[],"_revision_number":8},{"id":"d958fd0fd01b1c7a6542706bc5a3e8665009b080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 12:16:14.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6d957254072497d9e233087339d76a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9837cf56f24bb5925666072ea1115d : FAILURE in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7baa691acf554cac9fc81ef43542cdb8 : FAILURE in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0094d3a2fca94e24af29d838c17aba5a : FAILURE in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58f84a877e414e9f8ff30a15db36c4c7 : FAILURE in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a277d8a698474580c9d000eae7808c : FAILURE in 5m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/b0a70dae2936461f853b41ea626163e9 : FAILURE in 52m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b4f94e95ab6419d856b1cca6212bf3b : FAILURE in 25m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fb19147fa44426d9ec8869ca805b3ba : POST_FAILURE in 25m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e69f16538b8340dfb89a2fc676f9e329 : FAILURE in 5m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8a5cd9b8a7f2457a9cbbea692c5a6345 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e60e0de54702485eb57da9bec78e866d : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8e3d50d796474343938c28309c1a186f : FAILURE in 5m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/28541c5d4afa417a8d755ec8f0f501cb : FAILURE in 4m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/058e725903b44c8eb35152d546edb0a7 : FAILURE in 20m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96c984babfc8400d83b3844f3e9adb07 : FAILURE in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/adf6e751b0134980abbedc58cd34a12b : FAILURE in 20m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8bea0cc0a6ad4bc4bfc3d601fdea9078 : FAILURE in 32m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/81036577be8f4b5fa0ce1af6a7018fcc : FAILURE in 16m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a2a7c6384f2549438108fbccf19ffcdd : FAILURE in 24m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c4fe7ad77e074b7f96235c91e33b0145 : FAILURE in 27m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9de4f34118e74bf2b3c228dbec11742f : FAILURE in 1h 01m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/25a03d8786d245288b466db0933ce205 : FAILURE in 23m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4f74d87d6b5542158800b977a4ac62ab : FAILURE in 23m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/93a15bb344ae4b89afb73b5833bb2130 : FAILURE in 27m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8adb79b3b1fc40c88b33d2ad5a549273 : FAILURE in 38m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/13496dd464ec48569da9bef437d2891f : FAILURE in 34m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90a9faad17ac45499ade5ff25b1d0a28 : FAILURE in 29m 09s","accounts_in_message":[],"_revision_number":8},{"id":"8ba1b838309f38299e96d4cd9ed16f5a47cb469a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-19 12:24:37.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-5/FC/ : FAILURE in 5h 8m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-5/iSCSI/ : FAILURE in 4h 1m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"051dd66c26ee681bb51b507d9a97f71f40f3a526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-19 12:55:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"742a73af8c4dadb7f8205f750c6ac61ba10f5294","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-19 13:08:23.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/97/","accounts_in_message":[],"_revision_number":9},{"id":"4c1542b2d09aeb1b0764dc451226d3cb849eb82a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 13:08:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59a5a7089c4847b9bb55acb0e07a92e3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7196226261f41869cdb25d372aa6114 : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/362e8a88e3c742e088ecaf8307825ff3 : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a01a760a5147d58ab5426b2730978c4af0d3751e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-19 13:31:24.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/97/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"b5c7e623164d2ccc22c6666013bac86bcbea733b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-19 13:35:51.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d4e85ad46ed544aab92de91f5535bde2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/311b8e5f38a94ac7b94f28cf52636f75 : POST_FAILURE in 20m 35s","accounts_in_message":[],"_revision_number":9},{"id":"83869e8c325d5f06c4d2461245d1471a7fbb476c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-19 14:39:52.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 40m 49s","accounts_in_message":[],"_revision_number":9},{"id":"c2db7ccfe121014f66166e92f9b4b073e8d2720c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-19 14:41:41.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 35m 09s","accounts_in_message":[],"_revision_number":9},{"id":"14bebc7b12db8bc6b831c0fe003c58385c7a92fa","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-19 14:47:24.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 32m 10s","accounts_in_message":[],"_revision_number":9},{"id":"4f56365b3cd17e9a3a1d48d7cb82594143740fb6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-19 15:27:21.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-6/FC/ : FAILURE in 5h 21m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-6/iSCSI/ : FAILURE in 4h 49m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"bf7802c4dae3228ccd0bc8780e345e1225c7931e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 15:29:45.000000000","message":"Patch Set 9: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4bdf120bf85421491cdbf498b174e64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e33d5803084ea6808acd43cb701060 : FAILURE in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d1075c5bd8a4286a34e78212162f153 : SUCCESS in 11m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80a3de0550994168a30423eb4246c1a4 : SUCCESS in 10m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74d09e4e388a42a2b285f61df7a0bb8c : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b00f04cfb2d64b4fa1608de35845b428 : SUCCESS in 12m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/95022dadc21d4482829fd975edd17a4b : SUCCESS in 1h 06m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1cfd09a1775743128eeb3734063bf274 : FAILURE in 1h 15m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f940e3a8aa974f9c9acaf43493a4c2b5 : SUCCESS in 47m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eee15a7a586d4ad896acfa6f18bbf049 : SUCCESS in 16m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a0e17ae8b5554fa9a72c13385d96fae0 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ed80a7136b494d60a8a3e6d8dbc6ecf9 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/378d392ed87543619df7e6cfc605ff13 : FAILURE in 6m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ac0dd31ada584b1cbfdb946b18cb5e7f : FAILURE in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/29b4734c9f7745d196731d58a4c60342 : FAILURE in 28m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0b32033cb5214e2bb90fbd272f670f45 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61aa8c0071064428b42bee46f570d479 : FAILURE in 1h 58m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b3bb04266400490289d43658abcc1003 : FAILURE in 2h 32m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71ab848179d549ffae5381debf9f7160 : FAILURE in 1h 35m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/851fefe9cf704e70b6967156a9bcb478 : FAILURE in 1h 48m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3bab0e7025374d7db7f0a4f0e50963de : FAILURE in 35m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8f40cdf90bf84a3882215e252ba693f1 : SUCCESS in 1h 17m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/98db2100b4a547b0984bf156c6ff41d5 : FAILURE in 1h 31m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f7040cc73aac48c293fb221dfb5b4022 : SUCCESS in 48m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78d9e9886a984862b2e7028647993b30 : SUCCESS in 1h 23m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ddba97a4da6a432a9ba9658062ed2048 : POST_FAILURE in 2h 18m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca1003c49f734da8b8c1f246c2859548 : SUCCESS in 1h 50m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d5fd0bcf24d42f882c03e55fead8eff : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":9},{"id":"84174384eb909e0f4aec95da8dbb63ebd048d7f2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-19 16:10:45.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/16/918316/9/check/fujitsu-eternus-dx-iscsi/376971a : FAILURE in 2h 38m 54s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/16/918316/9/check/fujitsu-eternus-dx-fc/f2d1b39 : FAILURE in 2h 04m 28s","accounts_in_message":[],"_revision_number":9},{"id":"2429b779fff992801f8197450f1caa090b816662","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-19 16:14:17.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/760/.","accounts_in_message":[],"_revision_number":9},{"id":"c968ca02874c7b9834e3511ddc02be87337f7f53","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-19 16:51:42.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/27099532653a46b8881e389b790a197a : FAILURE in 1h 15m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5edf78cd536247419254db9e57784210 : FAILURE in 1h 06m 08s","accounts_in_message":[],"_revision_number":9},{"id":"ba5c5f7fde0a1d990cdd76eedf6d48fd9dbf8839","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-19 17:04:57.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/9/check/DellEMC_Unity_FC/89b8fc7/DellEMC_Unity_FC/2264 : FAILURE in 1h 30m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/9/check/DellEMC_Unity_iSCSI/894c2fb/DellEMC_Unity_iSCSI/2271 : FAILURE in 1h 48m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"849ebe60fcfa5c10ad54e2ce8cb02c91ce0d3ff0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-19 17:18:38.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/276/.","accounts_in_message":[],"_revision_number":9},{"id":"c6de5638427a27251f39153a805c8a09fdec2e8a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-19 17:21:31.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/9/check/DellEMC_PowerFlex_3.6/26d886a/DellEMC_PowerFlex_3.6/1752 : SUCCESS in 1h 47m 52s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/16/918316/9/check/DellEMC_PowerFlex_4.5/e03a0e7/DellEMC_PowerFlex_4.5/631 : SUCCESS in 1h 51m 04s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/10feec5/DellEMC_PowerFlex_4.x_NVMeTCP/589 : FAILURE in 13m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"1a82e35cce94e42f6e0625f8eb949c49156ac956","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-19 18:49:33.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-9/FC/ : UNSTABLE in 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-918316-9/iSCSI/ : FAILURE in 4h 13m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"a5fa8f4649ceef1aaed0f75947051ae81ebb2a9e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-20 00:48:53.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/9/20240720020101/NEC-V-ISCSI : FAILURE in 7h 19m 47s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/9/20240720020101/NEC-V-FC : FAILURE in 7h 47m 49s","accounts_in_message":[],"_revision_number":9},{"id":"2be239221710006796e7041d188defaeb119cf09","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-20 05:40:09.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/9/20240720051942/NEC-ISCSI : FAILURE in 8h 58m 35s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/9/20240720051942/NEC-FC : FAILURE in 9h 20m 24s","accounts_in_message":[],"_revision_number":9},{"id":"1c0f6551e20a23b53a528f0ed889c26e9d090f4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-22 09:12:48.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8754745472693e5086650e422334f4096e741b11","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-22 09:20:04.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/107/","accounts_in_message":[],"_revision_number":10},{"id":"f01003bdd139f292a179a32b7dedcfd69d13e070","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 09:35:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bf44c16ca924d35a62a1aacc6592e5d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51f6aaac30bf4bf5b91401d94d881af0 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04922f5c49b84907a2e6e5480f3d3701 : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"86d52df5420d797d3977406d634bb74def864bc2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-22 09:40:23.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/043e4f11ba2942f9953399b3ac8a63bc\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6a115551c3b0412ca975c69363d301a2 : POST_FAILURE in 19m 24s","accounts_in_message":[],"_revision_number":10},{"id":"88a18d11e62917103e38b182608120beb163549c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-22 09:43:04.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/107/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"b60737bf7f36a286bb51d6a86e1093b69beaf1a2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-22 09:46:48.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/16/918316/10/check/fujitsu-eternus-dx-iscsi/27d8a0b : ABORTED in 33m 10s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/16/918316/10/check/fujitsu-eternus-dx-fc/4408d60 : ABORTED in 33m 07s","accounts_in_message":[],"_revision_number":10},{"id":"7c6e693da3207e9728e5ddd82696a80619a1f06f","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-22 10:11:24.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 36m 22s","accounts_in_message":[],"_revision_number":10},{"id":"c7384b147309a77e7dd2b6e9814b72e64552a4a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-22 10:23:15.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 41m 39s","accounts_in_message":[],"_revision_number":10},{"id":"ebb98ef9993589a4e9a041c47e2e0f59a40b75fa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-22 10:26:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 41m 25s","accounts_in_message":[],"_revision_number":10},{"id":"9c727454995a65421e62c385c246ec1cf2edf5fb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-22 10:41:32.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/769/.","accounts_in_message":[],"_revision_number":10},{"id":"aea5d6207f84adb4947b6cc6b86a8a996814c732","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-22 10:56:07.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c26771758f1b4c878dbc7b562bd5b694 : FAILURE in 1h 20m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d7f7cd0bd2cf4bf183d9bd4d33dd2c4d : FAILURE in 1h 16m 34s","accounts_in_message":[],"_revision_number":10},{"id":"cbef4d3aa4920a65139966d49b2f580680678118","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-22 11:22:32.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/10/check/DellEMC_Unity_FC/2ad5349/DellEMC_Unity_FC/2273 : FAILURE in 1h 26m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/10/check/DellEMC_Unity_iSCSI/2121772/DellEMC_Unity_iSCSI/2280 : FAILURE in 1h 48m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"a20ddb1feb19823bab2eeb02466037bb8b6ed7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 11:23:37.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b58c101e496748139e2c39aa1c57dcdb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eac1d727e5847868683c46928ebfaf1 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d941bb4351ba49b5b0c9ce8d6e0a7b1f : SUCCESS in 11m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b0727dfb01d444b9ae4d3d86b45bfd4 : SUCCESS in 10m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6617678071c74cda8705bd353c3cc01a : SUCCESS in 16m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8c7dd0e107484e8b64a09887ef6fa9 : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/6600553e59e7441f952b9a2e9cae5063 : SUCCESS in 1h 02m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28a49bb88c62418ba0999a7fe431e5e3 : FAILURE in 1h 21m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6371ebdad514418882d59e0541a9b6cf : SUCCESS in 1h 01m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0d62a102fa0942aa8cc4e27e25824210 : SUCCESS in 17m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b2e71343b4044e3d97c546b04b20b6b7 : FAILURE in 4m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c65c155e61684e2693340108d9d06bf1 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/64f513cadfe040e6ae4fa2c8c4626754 : FAILURE in 8m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bcd356d479094d2d90e8f85978f06068 : FAILURE in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2ab0f35bef1480388aa7696f4ec7ccf : FAILURE in 35m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/905424b8738747de81241629b087f573 : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aaef850d8dcd4f0a93df4bfb02aa911c : FAILURE in 1h 29m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e44e5b9ae9804e7d831e90c00078b477 : FAILURE in 2h 10m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f84aff736074739983d944f611bd20d : FAILURE in 1h 52m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/efcd441f2e4e4593b2d31eba78238633 : FAILURE in 2h 04m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a77f2e2c8b4548868645e560dfabc8fd : FAILURE in 32m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/67b10974b62b4af48b9a5021b11ddb02 : SUCCESS in 1h 14m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2c4642afd1eb48bcbd0723004cac9214 : FAILURE in 1h 39m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/87890d71c999400a9afc27d99890d548 : SUCCESS in 38m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/291699e20f814550b65ba7c9a0b40a0b : SUCCESS in 1h 02m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/69ba1b96ef344f038aeaa8377b0f83c0 : SUCCESS in 1h 19m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56cbcea9b5c6449fb7a355aa7e79283d : SUCCESS in 1h 16m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ccee03ce3304c3389d497cc2a43415b : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":10},{"id":"e4ef11a6dff666b78a1520d4ab71b3bb81534fc8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-22 11:29:22.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/10/check/DellEMC_PowerFlex_3.6/a7e464b/DellEMC_PowerFlex_3.6/1761 : SUCCESS in 1h 53m 35s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/16/918316/10/check/DellEMC_PowerFlex_4.5/e9baa7e/DellEMC_PowerFlex_4.5/640 : FAILURE in 1h 55m 44s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/f7f9386/DellEMC_PowerFlex_4.x_NVMeTCP/598 : FAILURE in 13m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"4eb6be0e3d81cf8ccbc943c5c8939b315b8fa2e1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-22 22:24:14.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/10/20240722223554/NEC-V-ISCSI : FAILURE in 8h 01m 54s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/10/20240722223554/NEC-V-FC : FAILURE in 8h 48m 16s","accounts_in_message":[],"_revision_number":10},{"id":"ace88b74de156c2ca8f070bb550a26d6ce43f6d8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-22 22:40:53.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/10/20240722192450/NEC-FC : FAILURE in 8h 51m 31s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/10/20240722192450/NEC-ISCSI : FAILURE in 12h 15m 59s","accounts_in_message":[],"_revision_number":10},{"id":"82d6cec5c8da13192fc24d16885a2a3a3d4a111e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-23 11:31:07.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"e2a291ecad0d086f15c6de1669e30650b8f7f9a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 11:31:18.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/121/","accounts_in_message":[],"_revision_number":11},{"id":"5ff1235a680b8af73d7660076d7f85f978669773","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 11:31:20.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/121/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"be1ec215075e0755dcb1b0df6c85e1451b91e0c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 11:42:21.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d522c6c510431fab44d8b7c096b676\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/935cd4ddf9614004806c6bcd1c5a415a : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/19386fdbf99446c4bc44aa4184cc4fbc : SUCCESS in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e3964a1b90ad3895affa3d6a6faee892a3b0f2f4","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-23 12:53:47.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":11},{"id":"9ad5ef16bbc8b68d535cf1f048840aae8be6b148","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-23 13:07:00.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 36m 32s","accounts_in_message":[],"_revision_number":11},{"id":"cc51b6f9e2ded5dad8974aeb1166192d822615dd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-23 13:13:47.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 42m 41s","accounts_in_message":[],"_revision_number":11},{"id":"6c7dfe0c3dc01375f64d0ad94ae2dff2f5ed7301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 13:45:36.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9f5be8da94c445099ab2a61632f27c6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c461c7ace7374c51be6c81875f8a3c33 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596e5d65e7554cc9b8f6c099a413289e : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b96897c85e94c0e9b9e9e319ed2215e : SUCCESS in 11m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67c72836358342168fff94d2da599db4 : SUCCESS in 12m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa97fa89efda462d9f99035c0f3c319d : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51ca5c0585645a4bfca4e2f7a42f7af : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/942d52f8d9b84469a695c8331c2b10a4 : SUCCESS in 54m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e20aa3ebdeb341bebcdd76a99b0f33cd : FAILURE in 1h 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad5d2c625a104752a6299b6424dbd673 : SUCCESS in 45m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7624b08b30204266ad2833ea4768d56a : SUCCESS in 17m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0be59e53c5e64abdabe2ed86b11e3b58 : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04705cd791fe44aba423ebdc6b11f275 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d7dd4bed0bd427eb4292784377de6e6 : FAILURE in 5m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7a20c5e7d2fc4e31a49464f8329ce8a5 : FAILURE in 7m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4517988b719f4eeaa5b8e2e585ce6a8b : FAILURE in 32m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a519b93ada347a498255e7a4d051c61 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54ad1833214240a7ab22272641894050 : FAILURE in 1h 44m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bd8783464d39427bab714c5e0b63898d : FAILURE in 1h 55m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/31c75eea71c74ff6b81ea9555a2a9a80 : FAILURE in 2h 05m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fa474cecefad41b28f0c63eb1e2eec40 : FAILURE in 1h 44m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/70cfa07e36b7490c874f5f635b1bf1f0 : FAILURE in 34m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1d2def16d6954b1f82c6a5f2aaa0609a : SUCCESS in 1h 13m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0713b72ba08442d4bc6923e68ee17f7b : FAILURE in 1h 29m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4307d8161f0343a4852cc681c633505b : SUCCESS in 42m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/209e59fec2ec4704b35c780c2e6f759d : SUCCESS in 1h 16m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6bb4c6be39de4c49a01bb90044ba959d : SUCCESS in 1h 15m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d561f3cd03f040349af8bdaf7b3f90d1 : SUCCESS in 1h 35m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cde736950f1a438eb6626c24fe1c4a4c : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":11},{"id":"742164d29fe03bb7616fa155781aafdce016019a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-23 15:00:14.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/781/.","accounts_in_message":[],"_revision_number":11},{"id":"ad4dd9380d8a59765d9fb0917935d4db5c41087a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-23 15:28:27.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/43d4089dd1b64de7a55416d97b157562 : FAILURE in 1h 22m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9bc0dda860c94792bfb821dc9eebecf6 : FAILURE in 1h 16m 39s","accounts_in_message":[],"_revision_number":11},{"id":"aa6f17c0fe54158f295943f02d90c68cbc214ad8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-23 15:28:39.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4098ef0ea21348cc8b6a04c917fcf2c0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/850f9f5f77cb4b3fa4011200f4d11a6d : POST_FAILURE in 21m 11s","accounts_in_message":[],"_revision_number":11},{"id":"0bc5dc7a719db43a5ded0c82d71c4980a3e8ef4d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-23 18:32:37.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/11/check/DellEMC_Unity_FC/64927b5/DellEMC_Unity_FC/2285 : FAILURE in 1h 38m 30s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/11/check/DellEMC_Unity_iSCSI/77f7d4b/DellEMC_Unity_iSCSI/2292 : FAILURE in 1h 49m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"e98627cdf9dfc8f59b7066c93e9c7d39cd524e23","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-23 18:35:37.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/11/check/DellEMC_PowerFlex_3.6/2bf728f/DellEMC_PowerFlex_3.6/1773 : SUCCESS in 1h 50m 31s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/16/918316/11/check/DellEMC_PowerFlex_4.5/1ca3bf0/DellEMC_PowerFlex_4.5/652 : SUCCESS in 1h 52m 24s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/11/check/DellEMC_PowerFlex_4.x_NVMeTCP/6bc8348/DellEMC_PowerFlex_4.x_NVMeTCP/610 : FAILURE in 13m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"9e46850c91134811c67b50f55ab3ca656a1a0c41","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-24 08:00:07.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/11/20240724084020/NEC-V-ISCSI : FAILURE in 7h 29m 49s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/11/20240724084020/NEC-V-FC : FAILURE in 8h 19m 44s","accounts_in_message":[],"_revision_number":11},{"id":"f1e5daaeb014bea2daa2ef0fc5459bd3d72ba569","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-24 19:43:25.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/11/20240724191211/NEC-ISCSI : FAILURE in 9h 05m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/918316/11/20240724191211/NEC-FC : FAILURE in 9h 31m 11s","accounts_in_message":[],"_revision_number":11},{"id":"b3e9ae6aaa8bc00afa1b3720ff3061169c98d0a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-09-05 07:31:37.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"03a8b98b81f54678357d194cb2f626aa740497a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-05 07:31:46.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/485/","accounts_in_message":[],"_revision_number":12},{"id":"01be5f379dc4b95ae776be84d6a656076d3dc13a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-05 07:31:51.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"1c2f78cff493579151579c4f70287102b4416aaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 07:54:37.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/251bd0ee5a694c3e865fcfd506f8d6bd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8522699ecc8c43baa69ac3d990a5b231 : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a6a49c69375341cbae669d486e2347a3 : SUCCESS in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"865672a5e3897e13457bc268b44f5a7af60cf58d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-05 09:00:50.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3fcea1f9a0824d1bbb2fadf8a53e7537\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8422dfa8421f49af8d39eb4dca7e9d25 : FAILURE in 40m 02s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/389aaa5e084a4575bf1b0acbeb044fba : FAILURE in 1h 05m 59s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4df1b26210614252a2c67adc2614e254 : FAILURE in 1h 28m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8c9f14ee8ab0471bb7d057484eb0766c : FAILURE in 37m 33s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"7e740451765ff0cad84171336d1e173cd3d539c0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-05 09:08:30.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/488/.","accounts_in_message":[],"_revision_number":12},{"id":"c4710c7de1a066984cfe2e669c4db1523ab72268","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-05 09:19:59.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 38m 55s","accounts_in_message":[],"_revision_number":12},{"id":"9a3a621a094f3e108a3558556ee80467fb07f1fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 09:56:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66682f9ab6804b1482fe92ef807e39d7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/971344e6c68c4241a74eb4957ca4ba5f : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3dd18909ca84a268c3ae6aadf0d243d : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/443a93bb0c3f4247b4d159629aa305b2 : SUCCESS in 12m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f03b6b24ac324da299260f3d9679812a : SUCCESS in 12m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c741bdd098ee4eeea1e60def757bb141 : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd334337348345e594cfeaf9ab88f6c6 : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/49c6206cf7a84ddcaabd0bb72a4fc2b0 : SUCCESS in 55m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56ce76c06bf341f9a60e9fafcabe3cf2 : FAILURE in 1h 19m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60f725f682b04ced825200efa25276f6 : SUCCESS in 54m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c1181cfc4d3b418b94cc91a4f9a8a38a : SUCCESS in 15m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/07e29e87dab248fd9dffa7710d9f4543 : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b7a6d2abceb44395ae15b8be3ba19109 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/44828d2d6927438f9b6dc102d406ca52 : FAILURE in 6m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6d692d7515c34a20975fc2834dd2dad1 : FAILURE in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/72f230bf45154cae850643ded5dcc9d4 : FAILURE in 39m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e57cb50dda54451ba43b54bbd199b5d : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ba33aa56d038496994fe33c5f43582b0 : FAILURE in 1h 43m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1852d94f2d714c6b8aa8365ac814643f : FAILURE in 2h 18m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e5d71af295ea4bb181795fc8b8fc00b9 : FAILURE in 1h 51m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bf63b77d7aec417a8caccd0e8fcb4a33 : FAILURE in 2h 02m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a3105c4be2cf4901ac1c66251fe1ccb1 : FAILURE in 29m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88965a1a01004a54bff2c522ff5f6395 : SUCCESS in 1h 07m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97bbe4b6e3844c21af3b942c0d896a9d : FAILURE in 1h 39m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/545d975364e643ba8fcf9d04dbd310b0 : SUCCESS in 34m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/055af0910f164d4993a8ecab0c5bac97 : SUCCESS in 1h 36m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b28c6a5c6ef840cdad3f0971c719c0d9 : SUCCESS in 1h 21m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8e2bf7fa677244c7b53de54991374a58 : SUCCESS in 1h 33m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/743e115788c7491eb8c6cb581098d05f : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":12},{"id":"16f4e5d41f80bea8601f7337e3f6fc92265fd5e5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-05 10:09:42.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1076/.","accounts_in_message":[],"_revision_number":12},{"id":"12c19d0e7ffe9c4d37c80ffa7ecb7b6dce5d4f98","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-05 10:54:17.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2fe66197fe4a4037a6592888b42fe5ec : FAILURE in 1h 18m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/44f1ffd3da5e461d91ef9a40a1bf1552 : FAILURE in 1h 26m 28s","accounts_in_message":[],"_revision_number":12},{"id":"83e34c16d1a603443fe56b9fec11237279f99563","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-09-05 11:22:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/16/918316/12/check/fujitsu-eternus-dx-iscsi/00e8db4 : FAILURE in 2h 03m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/16/918316/12/check/fujitsu-eternus-dx-fc/97adf8d : FAILURE in 2h 06m 22s","accounts_in_message":[],"_revision_number":12},{"id":"4929b273e785a8d68b085044861dd7c9721c52ba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-05 12:25:16.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/12/check/DellEMC_Unity_FC/e101a1a/DellEMC_Unity_FC/2558 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/12/check/DellEMC_Unity_iSCSI/166699d/DellEMC_Unity_iSCSI/2563 : FAILURE in 2h 07m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"339195697cbba90a7464c827f4eafb140913bb56","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-05 15:16:38.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 35m 47s","accounts_in_message":[],"_revision_number":12},{"id":"448191a8ac97d5bd76b2e0a3e828d750c6a9b4e4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-07 00:00:57.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/12/20240907000413/NEC-V-ISCSI : FAILURE in 7h 56m 01s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/12/20240907000413/NEC-V-FC : FAILURE in 8h 56m 39s","accounts_in_message":[],"_revision_number":12},{"id":"0a5ec97a9db98034c3e8ab827726f9bb2f46c822","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-07 08:46:30.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/12/20240907080309/NEC-ISCSI : FAILURE in 8h 52m 03s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/12/20240907080309/NEC-FC : FAILURE in 9h 43m 18s","accounts_in_message":[],"_revision_number":12},{"id":"f52d5f2cf26f2f399ca1c58e1c1cb860919d5be4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-07 11:31:00.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/12/check/DellEMC_PowerFlex_3.6/b801385/DellEMC_PowerFlex_3.6/2054 : FAILURE in 2h 30m 49s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/16/918316/12/check/DellEMC_PowerFlex_4.x/0eeacd1/DellEMC_PowerFlex_4.x/937 : FAILURE in 2h 10m 46s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/76acc9a/DellEMC_PowerFlex_4.x_NVMeTCP/892 : FAILURE in 5h 54m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"190868b1d87bfd5708ffe94ca732603b72b51751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-10-08 11:51:36.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"bae8b467f6a80a28c5844abf89c5021019fbdd45","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-08 11:51:47.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/713/","accounts_in_message":[],"_revision_number":13},{"id":"29c59bf8986e4a3605df23dc75ba72bdcaf0b436","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37269,"deleted":true},"date":"2024-10-08 11:51:51.000000000","message":"Patch Set 13:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/39/","accounts_in_message":[],"_revision_number":13},{"id":"efd80929398943eaa1f404154b2b14e379e4cb22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-10-08 12:01:36.000000000","message":"Uploaded patch set 14: New patch set was added with same tree, parent tree, and commit message as Patch Set 13.","accounts_in_message":[],"_revision_number":14},{"id":"91d334e9dfbe811e914e4245353197db59f4efac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37269,"deleted":true},"date":"2024-10-08 12:01:49.000000000","message":"Patch Set 14:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/40/","accounts_in_message":[],"_revision_number":14},{"id":"b959f5fa93adf2506f6d8a7cc379c8d51c37c426","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-08 12:19:32.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 13m 40s","accounts_in_message":[],"_revision_number":14},{"id":"165edc4695fe1c1ea4865ee15cb086e2972f9146","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 12:23:55.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5adafde02ce4ff280a7792f14d0791e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8028af8df48b45308c2e66a4dfb44a85 : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/55211cd2cf064b58986eaaa684307c27 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a48edecbae4f28a79787a6c7f5d247e2977108ba","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-08 12:25:43.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 19m 49s","accounts_in_message":[],"_revision_number":14},{"id":"a33e0ab0908127b098beba2e3269078059e345fd","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-08 12:39:18.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 35m 55s","accounts_in_message":[],"_revision_number":14},{"id":"ea66b8ddd8cf764b1ed46db31f663b1c4328f365","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-08 12:47:38.000000000","message":"Patch Set 13:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/13/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"4c747eaaa708d55d7b61b6d6a801ffc8e2bd6951","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-08 12:47:40.000000000","message":"Patch Set 14:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/714/","accounts_in_message":[],"_revision_number":14},{"id":"87576d931bf7e51d4863e33858745d115778102a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-08 13:32:56.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4afcacf1eb4b4568bcffe46c980c0779 : FAILURE in 1h 28m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b6946f2523054a879c69bd90d8b29e4b : FAILURE in 1h 25m 53s","accounts_in_message":[],"_revision_number":14},{"id":"5c7f5514d24717655cc3ea30bc1a79a19119481b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-08 13:40:07.000000000","message":"Patch Set 14:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/14/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/14/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":14},{"id":"09b234e890088ce030e295e3d176085ead41d658","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-08 13:56:58.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5e521a91289b47848bd24232bf4283cd\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef498ecef6114a29ba70d742b3d3f762 : FAILURE in 37m 03s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0ec6308c6fb34febb5dc08c49d1f5fce : FAILURE in 38m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bbc090d3c6984bc8bb342734b3f784d3 : FAILURE in 1h 49m 54s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/36c7672be6c948ff88a3be88287804d8 : FAILURE in 36m 40s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b5b46bca397b4e51a3758b3766b9fdb2 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84ff3c4ff02547b6a17bdf0c8673e9c2 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a4ec176ef2f1421588f0232a40b4b61d : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c2517263e4b0487088ec5ee83bf45422 : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.ya... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"9431a8791161e5873d422674bad29ac004792752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 14:08:55.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8bed6681153488db8701f5c9bd6d8a3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/59d34ddf7a6149ffb1fb05597d0cdaab : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53acac1406c43cc967f4c88214ab0ce : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffb11f588a634373a22f941be8c856be : FAILURE in 10m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/742dda6bd4ad49f2b11b70a31b7cae4a : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96817f0b8ee14db1b3a217ab0e222381 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/5827e47f37e74914830e36372018159e : SUCCESS in 51m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de625e15128543d0badcd5900d0a59e0 : SUCCESS in 41m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f9ded771834487a8216d4c7318f73b6 : FAILURE in 1h 17m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b72d2c53f68c48a8a26ab3a43ff9994f : SUCCESS in 57m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/88af2d7c5a8c4d7ba1c458be7a4760c0 : SUCCESS in 16m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3daa078cf8154ffb98c67d250a1958eb : SUCCESS in 4m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ca511fc4eed74ae4ab4b05cf8fd66129 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31a76eb4e2a646029e4d773495c7649a : FAILURE in 6m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8f148416789a4d428eb4cc7f733868ef : FAILURE in 8m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b244c5adf30d44848dba7741935da765 : POST_FAILURE in 34m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59ccebd072e545c4ace54041393d168d : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/585fad982a604546ae10e91f0831552d : FAILURE in 2h 01m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7cdf2b891f52444984c735a83964c99a : FAILURE in 1h 36m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e891c973adf04fc0abb12ab006f939e8 : FAILURE in 1h 41m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bd0158fb3d8a4f70ab285814d6e63dd3 : FAILURE in 1h 54m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/72f2416319204220aac95fc06425a2df : FAILURE in 32m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2110f604f6a5425fad3845aeeebc3634 : SUCCESS in 1h 07m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/308053a5ea744d86892e9f2a23a66210 : FAILURE in 1h 20m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c92b084feca1491f88923fb44d63c603 : SUCCESS in 39m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4bfdc7a0920144d0be06149692c200d4 : SUCCESS in 1h 30m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a15a0db60f4e4a7f952ebb76c674d4df : SUCCESS in 1h 30m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c4b00138fc04d56ba254c18e2ddaf3e : SUCCESS in 1h 07m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5db34c806c674be58f7328d6e7baba6c : POST_FAILURE in 34m 59s","accounts_in_message":[],"_revision_number":14},{"id":"5c95e6a2f5ff50c203829b8785924be1e6f655a7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-10 13:24:39.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/14/check/DellEMC_PowerFlex_3.6/f068f50/DellEMC_PowerFlex_3.6/2217 : FAILURE in 2h 15m 21s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/16/918316/14/check/DellEMC_PowerFlex_4.x/a407162/DellEMC_PowerFlex_4.x/1100 : FAILURE in 2h 08m 38s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/14/check/DellEMC_PowerFlex_4.x_NVMeTCP/e3a4017/DellEMC_PowerFlex_4.x_NVMeTCP/1028 : FAILURE in 5h 52m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"ccf5da30edf92ddf9104e1693d463e1a1420b0ae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-11 07:48:25.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/14/check/DellEMC_Unity_FC/aaa4315/DellEMC_Unity_FC/2719 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/14/check/DellEMC_Unity_iSCSI/0505fa6/DellEMC_Unity_iSCSI/2718 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"b600049d11b31353c0e402da3933144c26f12e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-10-15 09:27:14.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"4a75285e547dcb090b5cc5c125ef3ef6bb355843","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-15 09:49:35.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 20m 20s","accounts_in_message":[],"_revision_number":15},{"id":"33d8858be61971c889108e281fdd1cce966f3276","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 09:54:34.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/003f8ea418d14628aaf010dfe41c66f9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9122d56df7494d31a711d90a9834897c : FAILURE in 25m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b21cfdd68d7f4817a4df0c84ea91589a : FAILURE in 23m 30s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2bcb3f0c9736d642b47661801f3e892387dc7dc8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-15 09:58:12.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":15},{"id":"ce194e107cde35ae4fd402d64445263ed2ca49e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 10:08:47.000000000","message":"Patch Set 15:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/738/","accounts_in_message":[],"_revision_number":15},{"id":"20c693729f4a9d84f81bea74a9e3f8b8ded7a529","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:00:26.000000000","message":"Patch Set 15:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/15/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/15/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":15},{"id":"cf09e87a4f4a8827860ebc26c76eaa7adab9fc8e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-15 11:29:57.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/430d1391298c400c93e74d8d8a78d0c7\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b3ff603230b34ff8bfec1b26ca840732 : FAILURE in 42m 10s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fbf467a9666240269a753ee7f49246c4 : FAILURE in 44m 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd8e88a6eac941ee94dea92579bb8e29 : FAILURE in 2h 02m 10s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cdd4216ce47848b4be082ee6be1a329f : FAILURE in 34m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e031436752e4dab988041bbc9325915 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dc61239607c2474bad4047a194e664b1 : RETRY_LIMIT in 4m 23s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d77893c359604a7aa8bfe0876d0d2430 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/52b80221cb7a472ba330d576c6ee6a9b : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul... (truncated)","accounts_in_message":[],"_revision_number":15},{"id":"6d5d28b83b5f713759597eb6103c979c82e0d3ef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-10-15 11:35:54.000000000","message":"Topic user-visible-metadata-in-volume-types2 removed","accounts_in_message":[],"_revision_number":15},{"id":"47135196327d73872ec0da40d9f4bae375a567b9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-10-15 11:36:00.000000000","message":"Topic set to user-visible-metadata-in-volume-types","accounts_in_message":[],"_revision_number":15},{"id":"f20fa75abe86685bc119bd5e00f190b4dd3ba18d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 11:36:11.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6485db56246447f9bba7fe1b91bdb08\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bcda071daede4a89a8609504411d3866 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fee2872ede534575ae9955f03f9594f4 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f68e18e1ed6d44b4b1cf446497995884 : SUCCESS in 12m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c17baee51e443439390e68f0a67b380 : FAILURE in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecfd8f3e31694b948335918b55c866a9 : SUCCESS in 15m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/7727f7b63c824baa9208a48687841cab : SUCCESS in 56m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6f740e995724e1bb652c18741bb085a : SUCCESS in 1h 00m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61993badbae44814aa19ea76ef9ecfc7 : FAILURE in 1h 15m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccdf6e8e86664966b5dbe57e8e2c3aa7 : SUCCESS in 57m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59bf55b3613548b883a742997ba6f314 : SUCCESS in 16m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/49c137e46a8145c3951761b3162494f1 : FAILURE in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/695cac639f684fcbb5119cf2d905f9e9 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f5e3503da8de434b996b7b401212072c : SUCCESS in 5m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/24a9c53ab17d4216a5323d7a411bf8e3 : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/729875d6bac84fbfb30d68eb652b62ea : FAILURE in 31m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4860e24995a84bc7a4b87ef661895626 : SUCCESS in 4m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/064f76f230474f72b0fea2baa987822b : FAILURE in 1h 09m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8f9ebcc5a06e4e8588de66b11c97ea33 : FAILURE in 2h 07m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eed4b316a98a4c369f83676a1338ff40 : FAILURE in 1h 45m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c1474e20001c405ab2c166d114a01f97 : FAILURE in 2h 01m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4d8a10efd655409a9dcf0d136f365e1c : FAILURE in 26m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/278e38c68b10442183701a5ee42b4c57 : SUCCESS in 1h 00m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4acafce2d1584f70a647bd19c3aa5250 : FAILURE in 1h 32m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b16be7715a4a4f7687fab6decd7cc094 : SUCCESS in 32m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0c4cf96c6f614101ab2c48b820c9dcf4 : SUCCESS in 1h 28m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/afa4ca8001b749eb9f18870752643f3d : SUCCESS in 1h 17m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b5bc2090a0be41dfad057a02a2b5df87 : SUCCESS in 1h 48m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f18ca4f5c4f74ec68e94bd95fec9872b : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":15},{"id":"98cae1ac511c5808af713d9a2df7768046ffb69d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-15 15:16:01.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/15/check/DellEMC_PowerFlex_3.6/dcbc347/DellEMC_PowerFlex_3.6/2239 : FAILURE in 2h 17m 15s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/16/918316/15/check/DellEMC_PowerFlex_4.x/fa6195c/DellEMC_PowerFlex_4.x/1122 : FAILURE in 2h 00m 57s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/15/check/DellEMC_PowerFlex_4.x_NVMeTCP/c938903/DellEMC_PowerFlex_4.x_NVMeTCP/1047 : FAILURE in 5h 48m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"3b237cb233796143013a14dd2f7ab4be79e6a9b2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-15 15:55:55.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d8ebccf50c574222819bbc075e498064 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e1f412c5b6504acbb7549210a9ed7c0a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"96ffa6f23f9c698c52a1897429a1759c2d3b11ec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-15 16:50:08.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/15/check/DellEMC_Unity_FC/207c8f7/DellEMC_Unity_FC/2740 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/15/check/DellEMC_Unity_iSCSI/a974567/DellEMC_Unity_iSCSI/2735 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"348fa0e63e90d09d038492dadf5fa5028a3f1667","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-15 19:07:33.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 31m 12s","accounts_in_message":[],"_revision_number":15},{"id":"953f1513d58ea60aebb0b2a8784a8ee799e97b25","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-16 08:09:49.000000000","message":"Patch Set 15:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/15/20241016074205/NEC-ISCSI : FAILURE in 9h 02m 52s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/15/20241016074205/NEC-FC : FAILURE in 9h 27m 40s","accounts_in_message":[],"_revision_number":15},{"id":"9a3383efc8c13b9197105ff0cf719539ed53ced5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-10-16 09:54:29.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"c4faec67b8c125d65da86651cd5c8c30c5a6c620","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-16 09:54:39.000000000","message":"Patch Set 16:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/754/","accounts_in_message":[],"_revision_number":16},{"id":"cc85326554f07e6c84fbdbe3dd4913391fbd01b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-16 10:00:11.000000000","message":"Patch Set 16:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/16/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/16/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":16},{"id":"309ba0875c0d52bb9de0370f8d4de6feb1649f6a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-16 10:19:29.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 22m 18s","accounts_in_message":[],"_revision_number":16},{"id":"3a49f836ae5155421107cf5d1eeeac88499c9c78","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-16 10:20:09.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 23m 46s","accounts_in_message":[],"_revision_number":16},{"id":"fd0260b90be1d6cafe7330e951469d78b8f223f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 10:24:24.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ff826bbc6674342be5e5384644f4773\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5390c051a2f4df888678f2dfb85f8cf : FAILURE in 26m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/afbde30b503e4cd5ab7adfdc61b48da4 : SUCCESS in 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"12eb0b17eda85e173cd7ea17a0cd6c15db675d17","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-16 11:18:55.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cc700e1501c24328afe470b7cb740a13\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3850b43b52ae40f8bb960a125a3cb4ea : FAILURE in 36m 46s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/26b087c3ef3242f2a62ec859aea09e5d : FAILURE in 1h 23m 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8677b093600a47eb98075657a6d11387 : FAILURE in 1h 23m 13s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd1d185cf0da44d59e5baeeb9c4ce4fb : FAILURE in 44m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dbdef078cd424ea38a5a6cb76619fc86 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0710fc4ff3cd483b8f73ee3d8a7dec30 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7cf79b5b4b874bb283ba5f336abc1c87 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9c1518537e0e4f2d81165ac48afb5771 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul... (truncated)","accounts_in_message":[],"_revision_number":16},{"id":"f3cc4a7bbccfcbd09b4ba00ac5494f88ce8c309c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-16 11:32:22.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4297c952509b458c871879da8a85bce8 : FAILURE in 1h 19m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/af77a4702dbb4f71b3d862ff7f493494 : FAILURE in 1h 28m 15s","accounts_in_message":[],"_revision_number":16},{"id":"df5973b50fbc565a9e6ca40bf8c787161530a491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 12:23:53.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ede27cd313144662bb0787215d429da7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/27813497222a406fa13b75e9b556d0b7 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7476f75db0da4bdc907f242afe79ee29 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19f752d6fb514dd78c67e1c5ebb975db : SUCCESS in 12m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/364fd9c30712482a8b32e3d6d1b0a5d8 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07cbc4678d0c4b5aa83c6c62c9145d42 : SUCCESS in 14m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/5c8b156db9a144ce81a239b4e5b1d4ff : SUCCESS in 1h 05m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c61b292d43b04c05afe0e64f6740f308 : SUCCESS in 1h 00m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/911e568a2eaf435aafbf32c78a1d0c4b : FAILURE in 1h 23m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/104ae44dcc9649c0be1764c0f8c9c4ba : SUCCESS in 49m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9bd93ab37e824d11a988f7da4d1f188c : FAILURE in 12m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/baf0e8e27a5c4f29bcc72cccc470a4a7 : FAILURE in 4m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d88d8f7187e8433ca0d0bcf949f2cd29 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb62908ad62e4ef9a62cf451b89fbf7a : SUCCESS in 5m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d14814be1be64d23b8c789a7882faace : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/806c02705a904660a45ccad162866d78 : FAILURE in 32m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b0d9ea7c504448d8dd0b88c9c63e112 : SUCCESS in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35927fb971ce41b5a672cec47b6b378e : FAILURE in 1h 39m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/23f024b05ace4b7996ed2da850658201 : FAILURE in 2h 28m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65682192fd99491297f621a118469743 : FAILURE in 1h 55m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/81daa9efa0f04d5a82fb18ffadbb7050 : FAILURE in 2h 19m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2508641c3e5b4d38bbd44682bcaf2494 : FAILURE in 24m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a3f516be08d44842b50a16db63fca253 : SUCCESS in 1h 16m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b7b188eff40145f0aa1172a7d490d481 : FAILURE in 1h 56m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/67f02827f5494975a0481b2aa1845cec : SUCCESS in 32m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f3dd2852975a4031bb599d3406312e65 : SUCCESS in 1h 26m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e379f07968274eba976241f8a0dcb744 : POST_FAILURE in 1h 42m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed213b7abe974b58952d1df3538d1c29 : SUCCESS in 1h 40m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b093dc64cbc410a8666b3b7ba91c9b2 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":16},{"id":"1a151757e46537fa5e33e205b52ce03c1fec4163","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-16 12:33:41.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 40m 38s","accounts_in_message":[],"_revision_number":16},{"id":"d59f9cd17fbb250f8f74fa95c55643aca811fa64","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-16 14:18:21.000000000","message":"Patch Set 15:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/15/20241016143048/NEC-V-ISCSI : FAILURE in 7h 56m 30s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/15/20241016143048/NEC-V-FC : FAILURE in 8h 47m 30s","accounts_in_message":[],"_revision_number":15},{"id":"5dbac838a3231c8223a4b2cf8c3478af02f47b46","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-17 14:27:25.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/16/check/DellEMC_PowerFlex_3.6/6e626a2/DellEMC_PowerFlex_3.6/2247 : FAILURE in 2h 12m 36s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/16/918316/16/check/DellEMC_PowerFlex_4.x/6b24f99/DellEMC_PowerFlex_4.x/1130 : FAILURE in 2h 03m 37s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/16/check/DellEMC_PowerFlex_4.x_NVMeTCP/fa10c60/DellEMC_PowerFlex_4.x_NVMeTCP/1055 : FAILURE in 5h 54m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"22e4cfaadee3df90f4ff254653e7f6aad1c96aa0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-17 16:50:50.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/16/check/DellEMC_Unity_FC/1a11609/DellEMC_Unity_FC/2748 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/16/check/DellEMC_Unity_iSCSI/bc3ce05/DellEMC_Unity_iSCSI/2743 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"aaebe9de58c0fde6ad3ab3e862b2192d22aa225a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-10-18 11:13:38.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"0efcc177904760befceeca4a9c32f8f3e24e0255","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-18 11:13:49.000000000","message":"Patch Set 17:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/766/","accounts_in_message":[],"_revision_number":17},{"id":"10844ecb98285c71107a13d1704f5265385e91c9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-18 11:14:06.000000000","message":"Patch Set 17:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":17},{"id":"cea4d7518e426e1d5393f5c1a485170bf4ac427b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-10-18 11:16:07.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"b0ece8d5763c06e1cfe4c66cc620c6ad04971fd7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-18 11:16:32.000000000","message":"Patch Set 18:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":18},{"id":"2b74d8064201079b4c581a20f35b071d2e699da0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-18 11:20:33.000000000","message":"Patch Set 17:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/17/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/17/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":17},{"id":"8fac2f622f0bb055fc28173bb4cff0b3931c3ad7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-18 11:20:35.000000000","message":"Patch Set 18:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/767/","accounts_in_message":[],"_revision_number":18},{"id":"d886b1ac1af4017f1c75d5132dfd8406b06d5cf8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-18 11:25:43.000000000","message":"Patch Set 18:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/18/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/18/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":18},{"id":"4b4dd82852dbf55676d2dd1251e478223dbc316e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:40:54.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60cb43f780d24e2f87aab36ec79ed98a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1161c5f8966343c1a3f05016e8c3de11 : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13767383d2bf41d2b018b2ecf7efbc97 : FAILURE in 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"d48f2b658e43a736558a2d750d9fb14588ded829","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-18 11:52:06.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 18m 56s","accounts_in_message":[],"_revision_number":18},{"id":"8641f69d28bc6e8d9d8b8ed353e1beada3d98901","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-18 11:53:20.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 25m 05s","accounts_in_message":[],"_revision_number":18},{"id":"bd0d5b3b728795bb2c0ea0b96185100396fb7476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 13:34:15.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cbe4d9f374446c489668974c6b74c6a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76cfc8271d54b6382602ac41a07ba19 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb8a9eb0191c4372afa705ce26e4a3dd : SUCCESS in 13m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6de0859bc2f4816a5375f11e1a5d154 : SUCCESS in 15m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ba850af88a454c93203c940a601db0 : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/802ce2faba4c4c698bebbef97c41df0b : SUCCESS in 38m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ff3251872f448ceb8412055baae7024 : SUCCESS in 38m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73d4786673554752b428f86bdee33a4c : SUCCESS in 1h 52m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd904ec35e824da6bdeaa392a5096c57 : SUCCESS in 40m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8260cd71c0ee45e594b0afee76fa7577 : SUCCESS in 15m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9ccd6d774a824ddd802bc880377fe278 : FAILURE in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7b731ea08cb4964a644f9934323fcbb : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/192d914d2a854c159ac0e5547c24bb07 : SUCCESS in 6m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d09c1ff9bb2648c4b18f69af3c7faf30 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5594f233989c4b7f9c56a37bc95eeae7 : FAILURE in 18m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7eb358e8536c45c39762665dc8dd3bd6 : SUCCESS in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/924fdc388ae44e15b2780afb80123b1c : SUCCESS in 1h 44m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f661c5ee9214475686f81557ebce1605 : FAILURE in 2h 00m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/354009d839d1468e8f9726f9e059291e : FAILURE in 1h 35m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6443f998af96498db0e0624daec48eaa : SUCCESS in 2h 17m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/91104987021a4a5aba4bf3f339639232 : SUCCESS in 40m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/acd2e8f2378a48d7bb13d5aee8c9fe3d : SUCCESS in 53m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c673fab9750345ecb98168120dbbb94e : SUCCESS in 1h 18m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4d9d2b4ffc1444c4bd30970c4a3ca95b : SUCCESS in 28m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ed339074e37b43689fc494f8b647d9cc : SUCCESS in 1h 13m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9b69d83a083e4df2bc6996752a5ef5d5 : SUCCESS in 1h 47m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/27c3c598b0c042f8984f812f302df6fc : SUCCESS in 1h 21m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02800fd35d5d41578c6f713586f58019 : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":18},{"id":"8ae3ca559ccd6f25c4170a9949dfd111ce994202","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-10-18 13:55:03.000000000","message":"Patch Set 18:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/718/.","accounts_in_message":[],"_revision_number":18},{"id":"b821c16f5ab96aefd843c52f5a1547dc93ee6267","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-18 14:32:28.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 36m 39s","accounts_in_message":[],"_revision_number":18},{"id":"290aeb574a1c5559746dc026d8c8184b2b9f1864","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-10-18 14:57:36.000000000","message":"Patch Set 18:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1292/.","accounts_in_message":[],"_revision_number":18},{"id":"236ed9008d89e705c94269a70bf8bd1df2394384","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-18 15:01:03.000000000","message":"Patch Set 18:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d12967fdf16f44a8a61486cdbee4af59\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f7bbb52197444ff09907b6874bc158a1 : FAILURE in 40m 35s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/247adb60a63145768ef03c43275595f7 : NODE_FAILURE Node request 200-0000010567 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de46625a9a8a4dcf80318e6476e2069f : FAILURE in 1h 43m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/049baeb597654d38970385c6e5e2ed6b : SUCCESS in 38m 15s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b14f5442a8c40239efa53a511d1c09d : NODE_FAILURE Node request 200-0000010570 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/408ef419c794408d97d8651790e0eebe : NODE_FAILURE Node request 200-0000010571 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c3a996ff63a4f499d234c83f580d708 : NODE_FAILURE Node request 200-0000010631 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e344f334f8e4ce1b667af1bc7d7250a : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job va... (truncated)","accounts_in_message":[],"_revision_number":18},{"id":"07b373a88ac4037a9456314e0a3966e05c29c2d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-18 17:19:41.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/653a414aa17b42d8b2bd24283af331ed : SUCCESS in 1h 10m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8767e4ae35434dc4b57a4b40e14b996b : SUCCESS in 1h 15m 42s","accounts_in_message":[],"_revision_number":18},{"id":"e51a2585c19264098e94dab956184c308347d42a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-19 18:09:52.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_PowerFlex_3.6/039a6f0/DellEMC_PowerFlex_3.6/2259 : SUCCESS in 2h 32m 19s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_PowerFlex_4.x/214337e/DellEMC_PowerFlex_4.x/1143 : SUCCESS in 1h 57m 59s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_PowerFlex_4.x_NVMeTCP/ecdcd96/DellEMC_PowerFlex_4.x_NVMeTCP/1064 : FAILURE in 5h 54m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"d3f3e27d7558b8daa5879a94cf1f604a86a3238e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-19 18:41:49.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/18/20241019232352/NEC-V-FC : SUCCESS in 3h 50m 55s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/18/20241019232352/NEC-V-ISCSI : SUCCESS in 4h 17m 53s","accounts_in_message":[],"_revision_number":18},{"id":"f5c4ac8136d56476c971b0b44fefddc4a2ddceec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-19 19:51:59.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_Unity_FC/45fea66/DellEMC_Unity_FC/2760 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_Unity_iSCSI/566955c/DellEMC_Unity_iSCSI/2751 : ABORTED in 6h 00m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"6af7353be84cb8c46ca82b78c4b63b995e0f1d74","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-20 12:48:02.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/18/20241020160718/NEC-ISCSI : SUCCESS in 5h 14m 29s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/18/20241020160718/NEC-FC : SUCCESS in 5h 40m 41s","accounts_in_message":[],"_revision_number":18},{"id":"7206cdd835e1ba29de0779c576a2ae9ea493c50a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-10-21 08:00:10.000000000","message":"Patch Set 18:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/16/918316/18/fc : SUCCESS in 0h 46m 42s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/16/918316/18/iscsi : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":18},{"id":"7b190dafaf6a5115f899e1ba2e7186a3ed827457","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-11-06 10:07:42.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"7b6468f7bb81f15b1ffddc60aaa85c0af6d34ccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 10:53:49.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee10d377fa2c4ad0b45063484c27c252\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0a2e0ac9fd34131be2e7c2f9606844b : SUCCESS in 18m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53c53659d924479e9c445cac4f89af3e : FAILURE in 32m 09s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"74a379d4836e31d6c564622c7253f7ab70e3471b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-06 12:37:32.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a45dcd0ad7b54a928d0784a7ec0bf4d0 : FAILURE in 1h 12m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a9b85fcd7dd149d0af30b5e61da91609 : RETRY_LIMIT in 8m 11s","accounts_in_message":[],"_revision_number":18},{"id":"4398624a5c88b796f8ee9a2869da39f07816f3b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 12:55:00.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a323c197374f7e992477c9e68590de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e80eca5aaa4c5292acf75f18c71f95 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b57f3323d8f48e3b9c3de3e50060a73 : SUCCESS in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18c45c9e7fc9406aaa2ac5da765d16a2 : SUCCESS in 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41869591983741b3877b14cc78bf5342 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/c00f6651d7cd4361ac3b05165f1fd9e0 : SUCCESS in 54m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7c3495f38ad248ab9f6e27f76b030a4f : SUCCESS in 39m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6bf3d86aa7b4a2cb8c6173cc119cc19 : SUCCESS in 1h 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cf21100807447bfbdcb14db1ae36693 : SUCCESS in 1h 00m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/78744f9ce6be48beb298e071e56e5c00 : SUCCESS in 14m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2d80274dcf0e494d876cb553d05273df : SUCCESS in 5m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5d5123880de44098a3616c32b484ba9 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae269882b5e64868b6cc76394dc35e50 : SUCCESS in 5m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c8d7b707ff944733bdbef6535e3dfb3a : SUCCESS in 5m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6ac76acccc70459da5730697851375da : FAILURE in 19m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c849f76d8678426f9a29b81cbb703fc7 : SUCCESS in 4m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/323a875a26124cacb156ffbc5339c8f5 : SUCCESS in 1h 10m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4fb4d10100d8402ab58d2e7769e2c2d6 : SUCCESS in 2h 30m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e0908ee0f1bf453fbd80d42808e431e0 : SUCCESS in 2h 02m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1808de7c2b134977b9f428d194da5bf3 : SUCCESS in 2h 31m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eb51128a59e5497dab6c393147c61a86 : SUCCESS in 31m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c561ee8b962d4e8c974330fa27220b22 : SUCCESS in 1h 06m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17022b2515424c65a7d6b714acb8bb69 : FAILURE in 1h 50m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9ae184e552104717b8359f4453e9de68 : SUCCESS in 41m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0061e8910db94b19aa62fb2062ff07f0 : SUCCESS in 1h 14m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e4161f35094c4b59b09e5da9423d94f2 : POST_FAILURE in 2h 37m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8093169325b141789e621bd100757978 : SUCCESS in 1h 20m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e12114c9ce0e4939aa3313675557be3f : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":18},{"id":"7c82a7a9c438cea803a0aa6e36623c0394a27ef5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-06 13:06:41.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/57f7264c4160413093672b3ccc209ac9 : RETRY_LIMIT in 5m 21s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/44bb8ca3082e42e5a1e22f302854aa97 : RETRY_LIMIT in 10m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":18},{"id":"66c2b4c6749391490bc6f6d3d1a4bd158ac4f9ad","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-06 16:30:27.000000000","message":"Patch Set 18:\n\nBuild failed.","accounts_in_message":[],"_revision_number":18},{"id":"5cd943cf31c8fe91474ab11021eb3a6ae5b6a344","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-07 20:26:32.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_PowerFlex_3.6/9bbff2f/DellEMC_PowerFlex_3.6/2392 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_PowerFlex_4.x/a244cb3/DellEMC_PowerFlex_4.x/1236 : SUCCESS in 2h 15m 47s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_PowerFlex_4.x_NVMeTCP/6eb180f/DellEMC_PowerFlex_4.x_NVMeTCP/1138 : FAILURE in 5h 53m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"180696b7f446b8b4b808b70884fa1d6e9450f22d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-07 21:44:12.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_Unity_FC/32e9995/DellEMC_Unity_FC/2844 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/918316/18/check/DellEMC_Unity_iSCSI/75c6b55/DellEMC_Unity_iSCSI/2819 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"8c0e5be1e1b64978ca39a8bd1397b35142e5dec0","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2024-11-12 13:36:14.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5a1a1c83438d61cd91ce02cdb1c845afe49c8983","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2024-12-18 08:04:36.000000000","message":"Patch Set 18: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"9b53b08882bcfe29ff8e4beb836b761449960eb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2025-01-06 11:44:07.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"80922c940e5ffe292bf797841caa474638ba92c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-06 11:44:17.000000000","message":"Patch Set 19:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1141/","accounts_in_message":[],"_revision_number":19},{"id":"2b879ad63d14951fa61d3352df5fe76aace310f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-06 11:45:10.000000000","message":"Patch Set 19:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/19/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/19/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":19},{"id":"8b4b22f924162df2c55aebaaa08d2d97545126e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2025-01-06 12:04:01.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":20},{"id":"f9e45f8ffe2f55c0176a6dbd920c8eb13f4f906b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-06 12:04:12.000000000","message":"Patch Set 20:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1142/","accounts_in_message":[],"_revision_number":20},{"id":"30117f973c3ba78bdf2e510dbcc5b3df29b9a896","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-06 12:05:06.000000000","message":"Patch Set 20:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/20/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/918316/20/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":20},{"id":"f03bb10385101faa8c73e15cadd68a8b3fc171d0","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2025-01-06 12:05:48.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"8dff65ed7e7a6368acaa7743184428b8b72033bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 12:23:33.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/927c5d73ffcf4de991af3b75498246af\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fdf25ee974c4958b786e848d51805f4 : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf7dcfe70b484128b2a18ed1de77fbcd : SUCCESS in 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"756a9d0a6d80ab891fdd51e925054426ec540b45","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-06 12:42:42.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918316 : FAILURE in 36m 46s","accounts_in_message":[],"_revision_number":20},{"id":"d6ab7f9751b28f0476eeec77a2215c339b29acd2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-06 12:45:35.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918316 : FAILURE in 37m 12s","accounts_in_message":[],"_revision_number":20},{"id":"34161cb1ad293e16aab32628c3282ebc8e8366fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-06 12:49:02.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918316 : FAILURE in 40m 28s","accounts_in_message":[],"_revision_number":20},{"id":"82b8258341bc92593847b5ff977e5b8d35c21605","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-06 12:58:03.000000000","message":"Patch Set 20:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/182/.","accounts_in_message":[],"_revision_number":20},{"id":"cb826f47cf6c381ff3e1ae70bf795f19cef283ac","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-06 13:08:19.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/16/918316/20/ : POST_FAILURE in 56m 44s","accounts_in_message":[],"_revision_number":20},{"id":"234c6ee59d1397d779892b5a5e8fc2a26d20c29c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-06 13:19:29.000000000","message":"Patch Set 20:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/171/.","accounts_in_message":[],"_revision_number":20},{"id":"8a84564e2a319130ddae196bbb34791551e92e56","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-06 13:29:40.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b881f4eb76c04faab298dbc8c08adc14 : SUCCESS in 1h 23m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7390afd8ce514338b7e2984b3a997cd9 : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":20},{"id":"a7101757491946ad88abe375479d6e5766cf649f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-06 13:39:36.000000000","message":"Patch Set 20:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3cb89ab9edd243bab6e611c1ab5b45c9\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7851ff159e9e4ca59d2f7f84fbad823c : SUCCESS in 38m 34s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1619fe4688ae43e8920154051f502d09 : FAILURE in 44m 07s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3395cfb690894f5abb4da2be8fd8c939 : FAILURE in 1h 35m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/28030f3515fe4c0e8c4419d9d14e9d9a : SUCCESS in 41m 38s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/69835ed3eead4997a9add92d8abf8dec : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1aa89aab29314345aa5ade908ca9fd8b : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5837902c04df499e9404128dfc264c97 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40fe097d5fc84319af5077f20fccc9ca : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":20},{"id":"bb41fc467d39e51578d9f96f36614e5237932931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:47:26.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55415abc76ae4c6486f29582a4adabdf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf0c0df9c8a49648f44fca41290605b : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35220b56f7a54923b3712a04253a9829 : SUCCESS in 11m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/330be1672ad54453b9a04ca22ef38644 : SUCCESS in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d4cdac26a5143d780f3f52cb1110d9b : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/4ed835f3145a4274847df9092556d9bf : SUCCESS in 36m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e3b71c4ed7a844e6a8d4b21fb10c2d46 : SUCCESS in 1h 04m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60a12e6f134f42338b1cce917b18935b : SUCCESS in 1h 38m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbddc515cec1403eabb2ad6ce20c0db0 : SUCCESS in 1h 01m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c18c5642e86b4ca3acc75e7a9396d7c4 : SUCCESS in 28m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f01fde4b93be46b1912a094801afd0d8 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f003bafff2d24bb69654fcfae4df8467 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce2a571239e34c2c8e895e50fca2cb78 : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/61f19fef99f5456c9d4a8d0b5f8eda72 : SUCCESS in 8m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f6883c7c7c464a22a28dbfe5c18477c4 : FAILURE in 37m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/57342fa6bf6c403194726e10ace46bda : SUCCESS in 4m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f61d399f33ee4f4a867655555d413897 : SUCCESS in 1h 57m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ee448aa3544c41788c2159d277309c86 : SUCCESS in 2h 28m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d9bb461b1e647a8806cb26954336be8 : SUCCESS in 1h 29m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/22dd0a0473ec425aa38e040963923169 : SUCCESS in 2h 05m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b1e65558269a435a8887344eafbea8de : SUCCESS in 22m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/72815fba6ffc491683a28e23cace3dd0 : SUCCESS in 1h 08m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97fd27df72194d3cb0757795ce4d3708 : SUCCESS in 1h 23m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dd145b8e7bdd4df896a0a64465667309 : SUCCESS in 49m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aa4381c0890b4810b0370136a7afdf49 : POST_FAILURE in 1h 59m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e6706e13da004d54b4af71b51aacefce : SUCCESS in 1h 26m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67a40af291e9430cb6a82af99ec7edc8 : SUCCESS in 1h 38m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a62d29f37264c7cb9558ca974adca85 : SUCCESS in 1h 30m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b9e6287b8144d56b911133163546c2d : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":20},{"id":"e4b1d3bfd0910e4d1be546d78076a810ce1af582","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-06 15:00:08.000000000","message":"Patch Set 20:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/16/918316/20/fc : SUCCESS in 0h 49m 05s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/16/918316/20/iscsi : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":20},{"id":"ced7cc8e67a4f8c75c794e689bb75492e7e666c2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-09 12:55:05.000000000","message":"Patch Set 20:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/20/20250109132915/NEC-V-ISCSI : FAILURE in 7h 32m 17s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/20/20250109132915/NEC-V-FC : FAILURE in 8h 25m 48s","accounts_in_message":[],"_revision_number":20},{"id":"437e6a06d7dd133ad4168aeb3aeb6646f335d902","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-09 22:12:23.000000000","message":"Patch Set 20:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/20/20250109214111/NEC-ISCSI : FAILURE in 9h 06m 45s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/918316/20/20250109214111/NEC-FC : FAILURE in 9h 31m 10s","accounts_in_message":[],"_revision_number":20},{"id":"f2fc8c128052bec45f92ca73e13970d03447314b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 12:46:30.000000000","message":"Patch Set 18:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2025_12_35_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":18}],"current_revision_number":20,"current_revision":"3511c9f36224a549ab7cff5a42feb93ac47dce3f","revisions":{"9d9b03b71a4a0e87321881e5a562d995ae5cf0d9":{"kind":"REWORK","_number":1,"created":"2024-05-06 13:14:47.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/1"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d9b03b71a4a0e87321881e5a562d995ae5cf0d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d9b03b71a4a0e87321881e5a562d995ae5cf0d9"}]},"branch":"refs/heads/master"},"14e0f2b73995510d7bd2b0eb84f3b71b92203c78":{"kind":"REWORK","_number":2,"created":"2024-05-07 07:27:29.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/2"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-07 07:24:36.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/14e0f2b73995510d7bd2b0eb84f3b71b92203c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/14e0f2b73995510d7bd2b0eb84f3b71b92203c78"}]},"branch":"refs/heads/master"},"85cebfa4775da4cff725e017e29d2bcba6691642":{"kind":"REWORK","_number":3,"created":"2024-07-05 07:26:02.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/3"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-05 07:24:49.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85cebfa4775da4cff725e017e29d2bcba6691642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85cebfa4775da4cff725e017e29d2bcba6691642"}]},"branch":"refs/heads/master"},"d75f457e0e68fbd6a31bca76143264b623ff73bf":{"kind":"REWORK","_number":4,"created":"2024-07-05 12:55:46.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/4"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-05 12:55:19.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d75f457e0e68fbd6a31bca76143264b623ff73bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d75f457e0e68fbd6a31bca76143264b623ff73bf"}]},"branch":"refs/heads/master"},"d4e653fb2b1ff54298a1262dfc397cc29e845985":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-19 07:04:34.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/5"}}},"commit":{"parents":[{"commit":"3d654c7f79abce32dbc4579b15edaea04447c362","subject":"Merge \"Remove cyclical import in dell_emc powerstore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d654c7f79abce32dbc4579b15edaea04447c362"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-19 07:04:34.000000000","tz":0},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4e653fb2b1ff54298a1262dfc397cc29e845985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4e653fb2b1ff54298a1262dfc397cc29e845985"}]},"branch":"refs/heads/master","description":"Rebase"},"cab2417cd4d4f0f7595a1d843dfb44c0410b3a75":{"kind":"REWORK","_number":6,"created":"2024-07-19 10:05:44.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/6"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-19 10:05:19.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cab2417cd4d4f0f7595a1d843dfb44c0410b3a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cab2417cd4d4f0f7595a1d843dfb44c0410b3a75"}]},"branch":"refs/heads/master"},"1bb00656eed24ca7de9f35868910d3c301d8114a":{"kind":"REWORK","_number":7,"created":"2024-07-19 11:00:50.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/7"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-19 11:00:27.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bb00656eed24ca7de9f35868910d3c301d8114a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bb00656eed24ca7de9f35868910d3c301d8114a"}]},"branch":"refs/heads/master"},"a0c5cf5ff7a2e18524bc68a10fe82b811d56868e":{"kind":"REWORK","_number":8,"created":"2024-07-19 11:06:42.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/8"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-19 11:06:25.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0c5cf5ff7a2e18524bc68a10fe82b811d56868e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0c5cf5ff7a2e18524bc68a10fe82b811d56868e"}]},"branch":"refs/heads/master"},"bd657d0b31152ed3fd2b147bfb1b6e2b0a279bac":{"kind":"REWORK","_number":9,"created":"2024-07-19 12:55:34.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/9"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-19 12:55:14.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd657d0b31152ed3fd2b147bfb1b6e2b0a279bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd657d0b31152ed3fd2b147bfb1b6e2b0a279bac"}]},"branch":"refs/heads/master"},"f16f39564f2c55584c4ff28abc0d4b48d796c508":{"kind":"REWORK","_number":10,"created":"2024-07-22 09:12:48.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/10"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-22 09:12:23.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f16f39564f2c55584c4ff28abc0d4b48d796c508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f16f39564f2c55584c4ff28abc0d4b48d796c508"}]},"branch":"refs/heads/master"},"8831400ebcfe6cfb64526ce92830f574eb826f5e":{"kind":"REWORK","_number":11,"created":"2024-07-23 11:31:07.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/11"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-23 11:30:41.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8831400ebcfe6cfb64526ce92830f574eb826f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8831400ebcfe6cfb64526ce92830f574eb826f5e"}]},"branch":"refs/heads/master"},"1dc754144395361c8294364cc5d69d89bbfc43ef":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-09-05 07:31:37.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/12"}}},"commit":{"parents":[{"commit":"25699617a1251d727a83e969fc27a16ce7b91f5b","subject":"Merge \"Dell PowerStore: Added timeout into rest API call.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25699617a1251d727a83e969fc27a16ce7b91f5b"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-09-05 07:31:37.000000000","tz":0},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1dc754144395361c8294364cc5d69d89bbfc43ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1dc754144395361c8294364cc5d69d89bbfc43ef"}]},"branch":"refs/heads/master","description":"Rebase"},"1db44e15b562dfda838bc6e619e3b5cd9f9bf07c":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-10-08 11:51:36.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/13"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-10-08 11:50:28.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1db44e15b562dfda838bc6e619e3b5cd9f9bf07c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1db44e15b562dfda838bc6e619e3b5cd9f9bf07c"}]},"branch":"refs/heads/master"},"0572028f14ae14a3c2accdb2291b7c1779b0d708":{"kind":"NO_CHANGE","_number":14,"created":"2024-10-08 12:01:36.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/14"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-10-08 12:00:59.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0572028f14ae14a3c2accdb2291b7c1779b0d708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0572028f14ae14a3c2accdb2291b7c1779b0d708"}]},"branch":"refs/heads/master"},"4935316af3721994296e0fa697a1f91867e84c77":{"kind":"REWORK","_number":15,"created":"2024-10-15 09:27:14.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/15"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-10-15 09:26:54.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4935316af3721994296e0fa697a1f91867e84c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4935316af3721994296e0fa697a1f91867e84c77"}]},"branch":"refs/heads/master"},"457f0efc1564e94fc41dda4163451e9ca97a7223":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-10-16 09:54:29.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/16"}}},"commit":{"parents":[{"commit":"8c5b35a1a96b10fa3af657e5ca83004278096ddf","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c5b35a1a96b10fa3af657e5ca83004278096ddf"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-10-16 09:54:29.000000000","tz":0},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/457f0efc1564e94fc41dda4163451e9ca97a7223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/457f0efc1564e94fc41dda4163451e9ca97a7223"}]},"branch":"refs/heads/master","description":"Rebase"},"d142cce95922c0846c15c98e8e19e3c00dae0fa7":{"kind":"REWORK","_number":17,"created":"2024-10-18 11:13:38.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/17"}}},"commit":{"parents":[{"commit":"8c5b35a1a96b10fa3af657e5ca83004278096ddf","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c5b35a1a96b10fa3af657e5ca83004278096ddf"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-10-18 11:13:16.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d142cce95922c0846c15c98e8e19e3c00dae0fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d142cce95922c0846c15c98e8e19e3c00dae0fa7"}]},"branch":"refs/heads/master"},"06879e4db26eef5444c9e3ba2fff2940d6e4cc70":{"kind":"REWORK","_number":18,"created":"2024-10-18 11:16:07.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/18"}}},"commit":{"parents":[{"commit":"8c5b35a1a96b10fa3af657e5ca83004278096ddf","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c5b35a1a96b10fa3af657e5ca83004278096ddf"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-10-18 11:15:50.000000000","tz":120},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06879e4db26eef5444c9e3ba2fff2940d6e4cc70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06879e4db26eef5444c9e3ba2fff2940d6e4cc70"}]},"branch":"refs/heads/master"},"cb043b538e40264918388c94a15e2eb55c900260":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-01-06 11:44:07.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/19"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2025-01-06 11:44:07.000000000","tz":0},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb043b538e40264918388c94a15e2eb55c900260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb043b538e40264918388c94a15e2eb55c900260"}]},"branch":"refs/heads/master","description":"Rebase"},"3511c9f36224a549ab7cff5a42feb93ac47dce3f":{"kind":"REWORK","_number":20,"created":"2025-01-06 12:04:01.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/16/918316/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/918316/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/918316/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/918316/20"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-05-06 12:58:43.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2025-01-06 12:01:23.000000000","tz":60},"subject":"Add volume type metadata table to database","message":"Add volume type metadata table to database\n\nFor the volume type metadata there is an extra table needed.\nThis commit adds the database table \u0027volume_type_metadata\u0027.\n\nPartial-Implements: blueprint user-visible-information-in-volume-types\nChange-Id: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3511c9f36224a549ab7cff5a42feb93ac47dce3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3511c9f36224a549ab7cff5a42feb93ac47dce3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
