)]}'
{"id":"openstack%2Fcinder~918434","triplet_id":"openstack%2Fcinder~master~I2cbae1b6759da8271b481541bf5ba7647de3498f","project":"openstack/cinder","branch":"master","topic":"remove-urllib2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-18 16:41:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2cbae1b6759da8271b481541bf5ba7647de3498f","subject":"Remove fallback for Python 2","status":"MERGED","created":"2024-05-07 12:09:22.000000000","updated":"2024-10-18 16:42:31.000000000","submitted":"2024-10-18 16:41:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918434-remove-urllib2","meta_rev_id":"3f6adec09f23a493278d277076a23f871abb5dab","_number":918434,"virtual_id_number":918434,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-18 16:41:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2024-05-18 02:56:27.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2024-10-18 14:43:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2024-10-18 14:43:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-07 12:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-07 12:35:04.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-07 13:16:35.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-07 13:37:49.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-07 14:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 15:47:53.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-07 17:34:48.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-05-07 17:43:21.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-07 19:16:52.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-08 17:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-05-09 02:46:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 12:23:00.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-05-13 02:58:13.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-18 02:56:27.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-10-18 14:43:01.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"ae6cbc8c8ac8bdbcade52afb27100e2889b83c23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 12:09:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ddb7edff8b0615425dd93c07b72d34fa63f0ec2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 12:14:23.000000000","message":"Topic set to remove-urllib2","accounts_in_message":[],"_revision_number":1},{"id":"4a06d91ca0e02d8fe19c2fee279ff7034e9867ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 12:20:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31518d78375b4cda8d0fb07b8e6e9b60\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3d7d45182d745bc9d2800d14b749db7 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/90ddfc4fc41548aa8551f2156d56c90c : SUCCESS in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59952ad406888cc4d48709769544a2494d334b28","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-07 12:35:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918434 : FAILURE in 24m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c8e33be134667795661d44973fef79af6ce4ceeb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-07 13:02:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918434 : FAILURE in 23m 44s","accounts_in_message":[],"_revision_number":1},{"id":"af2e233bf266a29cc6f7f2226963b6e46d72bb9e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-07 13:16:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/31/.","accounts_in_message":[],"_revision_number":1},{"id":"e1a83eaa53e3200ec3be7e70b60d0ee784a6eb4e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-07 13:31:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/342/.","accounts_in_message":[],"_revision_number":1},{"id":"edaed323c6e72af6c8dcbef5435c40c1b385d2cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-07 13:37:49.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/bc3975bd48764bbb961f2e13d3d1f154\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/183d24c1d5384132958b3915fa402df5 : FAILURE in 1h 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"38f25708a81331bf9d140d3147d9b1486a2a7e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 14:44:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f162538825004f679bb765abadc46301\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee3af2e8c4f42688650c3af1c3ebbd1 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b4eb9dcdceb47008559e4ad654cb440 : SUCCESS in 10m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b1df63597d34f8d8feeb069c14c6c6e : SUCCESS in 10m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd40db7ca44d423c98ffc7e71678e658 : FAILURE in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61018698d8a4a9ab3ceb144b29aa171 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/c67e58e9952f46b59839f361187c747f : SUCCESS in 51m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5b638707fe14e84a32adc3c186f910f : SUCCESS in 1h 16m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93a86f2c7df44c24bd4b6e5ecddee6db : SUCCESS in 1h 06m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f91bbfcefde44caaf511e921c095d81 : SUCCESS in 14m 11s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6fae264d2970498c953b20bd929224f3 : SUCCESS in 13m 15s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eccb17df21954a5989cd50c28d659e94 : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/344bb279e19843c88c01630ac7b5fbe0 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f09c499c832e4dc7b9e0095fafbe4bd7 : SUCCESS in 5m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4dbf0135dff847e2bcc534d96cfa1d5b : SUCCESS in 6m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/132ee351dcc84a32a2ab415e98f2dcda : SUCCESS in 49m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/25c82507847542d1bf3e5b6aca87c126 : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/39dee570c8d044cca43c64e76ce8d575 : SUCCESS in 1h 51m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b95cc0ed645a4009b8f35a51183172df : SUCCESS in 2h 15m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5b4f6c29f7a04affa4f585a84e7a6fb3 : FAILURE in 1h 53m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b3cd9feb3d4549618fe61eefdb2bc89f : SUCCESS in 1h 52m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8eb36a2b20c64de8b14116db9ee893d8 : SUCCESS in 28m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9d48ff737f5f4c0592f7b0cf0ce2a0b2 : SUCCESS in 1h 04m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5210c4314bde4421aa4775ed7c657259 : SUCCESS in 1h 44m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/30f5c95cdb824f2da704e314a34ae3ac : SUCCESS in 51m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cb65d4d2ff774606baea5997474a6a16 : SUCCESS in 1h 06m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/17bef692a3594545ac8b316baa3a17de : SUCCESS in 1h 12m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/96346dece38142f292bd9f4e6e6a04c6 : SUCCESS in 1h 24m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8f1b839ee6a4707bb7f27b7d1e00763 : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5a950c2290df1ad9b4bd6a0b5bcb69c1ac443fd5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 15:11:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53094b3350b7645217906b003df40c3596e588ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 15:34:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c56d0ec79164298b87be49582ccab8e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8515b208cf944ffc98db6a3ae588db1d : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c884631938b54ae3836a9edc6ae9eeea : SUCCESS in 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27a732c73c28e621be4236e49084d81b31d9ef92","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-07 15:47:53.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_Unity_FC/1627b6e/DellEMC_Unity_FC/1820 : SUCCESS in 1h 22m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_Unity_iSCSI/a24a836/DellEMC_Unity_iSCSI/1848 : SUCCESS in 1h 41m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d6215d8d5fdb1b01735849ace4e25c0be65105b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 17:28:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9af8c7e02ec47be89ea2db64aaa316f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19e3b4adbb34d3b9397308da54f5c9f : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13b9eeb5623a4545b11b5e20aff8227a : SUCCESS in 9m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70e537d1a9e94995a0e520b30bcc3dbe : SUCCESS in 12m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54ca5136b9484b51abf6c2c78e94a0e8 : FAILURE in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78759418025a4a839abb94ba680ba057 : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/3a58b416d8324cd09e873987f5d867e2 : SUCCESS in 1h 02m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/605249e6737f4b6ebd6e3495379131f1 : SUCCESS in 1h 37m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a555dc3ca8e541719470849ffbf38fdb : SUCCESS in 49m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/58f6a469f07642d4b9e0db8377a819b1 : SUCCESS in 18m 00s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d99f1735af574740bb7088488df7a7f8 : SUCCESS in 12m 30s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7aee3881cc64f90a2815cefe6a07748 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20ee49fe8ddd450f955a55fe870708f1 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e9644e41eed44258b46ef0e658dcde68 : SUCCESS in 6m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/52199633b9514db0aa6e3b9db94960ca : SUCCESS in 7m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82b65a0d087942bba25e27e2501dd8e3 : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e4ca4f641bf349a0a799f04c23ed9431 : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0f598e4291d7413e849aa45252121e00 : SUCCESS in 2h 06m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b01035549b2849d0a4b4b87bc9f9a0a8 : SUCCESS in 1h 55m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3fede098a87a4e229c5ce9cf524b6f71 : SUCCESS in 1h 53m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2b12d99caf6a4f85b7122b341f050cf9 : SUCCESS in 1h 50m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/31ea6ba93a1e4348b56ca374db793085 : SUCCESS in 37m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9fe92c506bad46169c51cf2254f0bfd1 : SUCCESS in 1h 12m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fac6f4af89764eeea2be0b309919b200 : SUCCESS in 1h 23m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0d6c6030ca0244299c54b40fcd374763 : SUCCESS in 32m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5c5f88459567465c8e5fcf6976871c32 : SUCCESS in 1h 22m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d5516f3d122f43e68ddd127a4d797076 : SUCCESS in 1h 23m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d810886fc74c46e5aa489d08581836bc : SUCCESS in 1h 32m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2144ce5d3561476393a43de97a5277c8 : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a5226499779317d2c6998ee8a4936aa0caa6b82d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-07 17:31:17.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9ffafca3a43a482695747ec1cb2c741d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/15814e7cd6134a6cacc2ec06090f7ee2 : FAILURE in 1h 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7bde86f14635701feddb17ddcc5652cfd9e96606","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-05-07 17:34:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_05_2024_17_28_42\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"d10bc9171be58899757765b8d6ef9dec64a4ca5f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-07 17:43:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/918434/1/upstream-check/cinder-tempest-plugin-ontap-fcp/f9c9d27/ : RETRY_LIMIT in 3m 42s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/918434/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/2c69fd6/ : RETRY_LIMIT in 3m 35s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/918434/1/upstream-check/cinder-tempest-plugin-ontap-nfs/24ef3ce/ : RETRY_LIMIT in 3m 10s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/918434/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/7ac2ea4/ : RETRY_LIMIT in 3m 44s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2dfe6a3a443c1743f771c9b8a61efa50fc276959","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-07 19:12:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_Unity_FC/d56c19a/DellEMC_Unity_FC/1822 : FAILURE in 1h 42m 20s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_Unity_iSCSI/f1b21df/DellEMC_Unity_iSCSI/1850 : SUCCESS in 1h 43m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"906f6c478fdf9496f7eb1b947b197490ce493ca5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-07 19:16:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_PowerStore_iSCSI/3a26c3e/DellEMC_PowerStore_iSCSI/1301 : FAILURE in 1h 23m 45s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_PowerStore_FC/4b4c9ea/DellEMC_PowerStore_FC/1471 : SUCCESS in 1h 00m 09s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_PowerStore_NFS/c49fc54/DellEMC_PowerStore_NFS/991 : SUCCESS in 1h 09m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_PowerStore_NVMeTCP/cc7fe41/DellEMC_PowerStore_NVMeTCP/851 : SUCCESS in 1h 48m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"c4f2b448d8f9f1b1f7d1130fe7f5ac00e15a0e63","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-08 17:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/918434/1/fc : SUCCESS in 0h 43m 31s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/34/918434/1/iscsi : SUCCESS in 0h 47m 38s","accounts_in_message":[],"_revision_number":1},{"id":"279f54232198947e40d343c81eabb0729151862d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 02:46:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/918434/1/20240509070208/NEC-ISCSI : SUCCESS in 4h 20m 33s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/918434/1/20240509070208/NEC-FC : SUCCESS in 4h 44m 11s","accounts_in_message":[],"_revision_number":1},{"id":"35c4b43c266c86b5bd2c44c95208e2af0ac77cb5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-09 12:23:00.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/918434/1/20240509100339/NEC-V-ISCSI : FAILURE in 9h 56m 13s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/918434/1/20240509100339/NEC-V-FC : SUCCESS in 11h 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"baa79f52bc65004cd4f8ff11e7f438d246ef032f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-13 02:58:13.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_PowerFlex_3.6/35f33ec/DellEMC_PowerFlex_3.6/1306 : SUCCESS in 1h 44m 18s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_PowerFlex_4.5/530445d/DellEMC_PowerFlex_4.5/232 : SUCCESS in 1h 52m 45s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/34/918434/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/432b7c8/DellEMC_PowerFlex_4.x_NVMeTCP/178 : FAILURE in 2h 54m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"ec6f9f47b85bd5e965d0651e335fe497c69fa763","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-05-18 02:56:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"20b80d99512699972f9edfe7ed07b03a6b5b7684","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-20 10:44:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/53/.","accounts_in_message":[],"_revision_number":1},{"id":"d5cad47339c2ae685d9a15c3865b90bd870fac6c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-10-18 14:43:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3db0ef9f991e35d1a881bcfc7a3816161cc9c60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 14:43:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1b98da61837ff95f1d55b383f90d5bd123f9eee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:41:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/169bfe81bd444fd38218a94bc28e202f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f139de6f58654d2ab035cdf1c9325fb3 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b2e962d8837499c9d8d88f931a3b95b : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a9a75aebd9641979035e45b4648893e : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b5edb060224ffbb2cf24f7750e8b33 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/016d732a12ad49e8b722d1b71b6dbe66 : SUCCESS in 57m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8bd012be40ef4bb0b227c0601fa6c450 : SUCCESS in 59m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7b490a964a4442a8ae117a13c6fa4ef : SUCCESS in 1h 21m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/faea6c71ddfe48ecaa6db8d1459c2c3e : SUCCESS in 57m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cb476e8bb4a64ff69cae5e4ab468c7d4 : SUCCESS in 1h 10m 28s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a046692bc67041a4920a5e97213ad3c6 : SUCCESS in 1h 51m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/926f1a468a714f269765d90cb2d81db7 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c453ca54d6b67aa66f4d79ea99a8c8134084efb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:41:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3f6adec09f23a493278d277076a23f871abb5dab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:42:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dd323e14dde4c0fb6eaf8ca4d6bd5c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7fa64c5843468d8158026cbf713b9c : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7dae0e5bdf3151b60e88ee6b5f77b335ed76f026","revisions":{"7dae0e5bdf3151b60e88ee6b5f77b335ed76f026":{"kind":"REWORK","_number":1,"created":"2024-05-07 12:09:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/918434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/918434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/918434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/918434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/918434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/918434/1"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 12:05:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 12:09:11.000000000","tz":540},"subject":"Remove fallback for Python 2","message":"Remove fallback for Python 2\n\nPython 2 is no longer supported, thus fallback to urllib2 is no longer\nused.\n\nChange-Id: I2cbae1b6759da8271b481541bf5ba7647de3498f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7dae0e5bdf3151b60e88ee6b5f77b335ed76f026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7dae0e5bdf3151b60e88ee6b5f77b335ed76f026"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
