)]}'
{"id":"openstack%2Fcinder~918805","triplet_id":"openstack%2Fcinder~master~I4116412db30602260049a4129cdfbbc5d5c20c73","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-05-20 08:38:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4116412db30602260049a4129cdfbbc5d5c20c73","subject":"ceph backup: Only warn about striping opts when necessary","status":"MERGED","created":"2024-05-09 17:09:54.000000000","updated":"2024-05-20 08:42:10.000000000","submitted":"2024-05-20 08:38:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918805","meta_rev_id":"91c9821394e0940261642cc903a5ac53aa74d1eb","_number":918805,"virtual_id_number":918805,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-20 08:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-05-17 15:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-20 04:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-20 04:36:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-09 17:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 18:42:31.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-09 18:45:19.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-09 19:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-09 19:32:33.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-05-09 19:36:35.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-09 21:02:56.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-09 21:29:47.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-10 03:13:51.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-05-10 23:03:24.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-05-11 19:27:37.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-13 09:13:00.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-05-13 22:46:33.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-14 02:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-05-17 15:00:04.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-05-20 04:36:21.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"2eac9443119cc93f2da8b8927157ea49bf05c95c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-09 17:09:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0d66cd8a73422184eb76494b46ef3378614e5d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 17:19:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbfaf711a18f4634961962474f33835d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/494134540f134f73bfe3e9d799ded9ff : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a378b263c5b40859b0a5592055204a7 : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"daeba7055ef4bbe418b887ce5273a065a24b8d11","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-09 18:42:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/350/.","accounts_in_message":[],"_revision_number":1},{"id":"14d56d1aa3f925e9717c1a75943f528dc54faf8d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-09 18:45:19.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4a04ae11236b425a82fd4e9cf0ec7172\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8693d928524e43909159fd586a10f2d2 : FAILURE in 1h 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"590c04370fa49d7e188adc47f6ed14b0150ec582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 19:26:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae13ea8f55e4d32bde1bba566470676\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5874b9810bb4dca8807b2ced8f3f3fe : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e3a25488d90481698036059bbfc2d18 : SUCCESS in 10m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/771b1155da4f4e5a855b633ac96f5012 : SUCCESS in 9m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dbc315670c6e4a83a43e102091cca5d6 : FAILURE in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3617aa61156e4c54833aba1a13ce7a12 : SUCCESS in 14m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/fe6a70bcbdf04f4db45f031c36758135 : SUCCESS in 50m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56e78b4fe96c456990be287bcacab728 : SUCCESS in 1h 45m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcdbb97d0cf74bb6a785772a8abdcfac : SUCCESS in 1h 01m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2bd1c095eb0642b9958b5e64fad4cae0 : SUCCESS in 17m 04s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/54b730baa1d24fc4bcaf38423821693b : SUCCESS in 13m 38s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f28f6d514464404833099df42e54439 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83e03391d25c4deb9361a31f11c5f368 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8197ed716611498d851d6db9b52dd3f8 : SUCCESS in 6m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68222ba3a6994c72b7df640c20b86d62 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e04aa101e68d45728bc33abbc09ae51f : SUCCESS in 37m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7b5e16b079045928b5510c736c29bc3 : SUCCESS in 6m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eaf393917201475f803a26650af19ae7 : SUCCESS in 1h 43m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e87e92c49ff47c5ae9d9eb4c45b3c4c : SUCCESS in 2h 09m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f526edc5cfad4d2ab8632f34145040e3 : SUCCESS in 1h 49m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3e159cfc7417413ea1213a91d9d14150 : SUCCESS in 1h 55m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5f2cf7516c48467d90b2389f1500f9ba : SUCCESS in 37m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9fd44f84c8b94a289123439c3201a3fe : SUCCESS in 1h 19m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/40e9efb17a6044e689a7fe926a5c0d46 : SUCCESS in 1h 20m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3fdca76f9ae34dd28c8af84b427d74b2 : SUCCESS in 46m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aaa97be777754600972f7c45cae8d55b : SUCCESS in 1h 14m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a3d9fb6d256846c2941ec78105a661ae : SUCCESS in 1h 14m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9316775dd004611806dba581f42e725 : SUCCESS in 1h 38m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9a7a68412c3496eadd103f6f4fec743 : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":1},{"id":"78c66a76885dfbb6bb87823afda2aecc3f16a666","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-05-09 19:32:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_05_2024_19_26_26\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"0d5466dc2bdfdeccd419631c6e0b0d3ea9f29865","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-09 19:36:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/918805/1/upstream-check/cinder-tempest-plugin-ontap-fcp/059553c/ : RETRY_LIMIT in 3m 55s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/918805/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/46f1659/ : RETRY_LIMIT in 3m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/918805/1/upstream-check/cinder-tempest-plugin-ontap-nfs/cf94547/ : RETRY_LIMIT in 3m 59s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/918805/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/86f425e/ : RETRY_LIMIT in 4m 01s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4a29581161bb0c27518cac1d1e4430a39dcefa3f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-09 21:02:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_Unity_FC/1895187/DellEMC_Unity_FC/1828 : FAILURE in 1h 37m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_Unity_iSCSI/6cb19ca/DellEMC_Unity_iSCSI/1856 : FAILURE in 1h 49m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1b73c94c34e2f51d769e9fd7c1fbe64564cffbe7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-09 21:29:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_PowerStore_iSCSI/2e9fe15/DellEMC_PowerStore_iSCSI/1307 : SUCCESS in 1h 10m 34s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_PowerStore_FC/1990043/DellEMC_PowerStore_FC/1477 : SUCCESS in 53m 51s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_PowerStore_NFS/6563be8/DellEMC_PowerStore_NFS/997 : SUCCESS in 1h 08m 18s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_PowerStore_NVMeTCP/ea72133/DellEMC_PowerStore_NVMeTCP/857 : SUCCESS in 1h 47m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"0b66fdafa4ab9035ebd324718d4b7405a14e6f5b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-10 03:13:51.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-918805-1/FC/ : FAILURE in 9h 18m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-918805-1/iSCSI/ : SUCCESS in 3h 6m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"095a65f9e8ea15c0c8692ecb914bfd0022a40e8f","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-05-10 23:03:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/918805-1/HPMSA-iSCSI/ : SUCCESS in 3:05:31\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/918805-1/HPMSA-FC/ : SUCCESS in 1:38:50","accounts_in_message":[],"_revision_number":1},{"id":"ca995a3e799d56846cce3e8ea8afb371be2a2e96","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-11 19:27:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/918805/1/20240511193453/NEC-FC : SUCCESS in 4h 50m 39s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/918805/1/20240511193453/NEC-ISCSI : SUCCESS in 8h 52m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e1891424bedf85afa5367d644c27fd82ae6d5426","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-13 09:13:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/918805/1/20240513150310/NEC-V-ISCSI : SUCCESS in 3h 05m 51s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/918805/1/20240513150310/NEC-V-FC : SUCCESS in 3h 09m 47s","accounts_in_message":[],"_revision_number":1},{"id":"546d5397c26e94a8582f4392d17aa483e21e448b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-13 22:46:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_PowerFlex_3.6/85aafa5/DellEMC_PowerFlex_3.6/1313 : SUCCESS in 1h 30m 15s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_PowerFlex_4.5/7e2fd34/DellEMC_PowerFlex_4.5/240 : FAILURE in 2h 19m 59s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/918805/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/7755f4d/DellEMC_PowerFlex_4.x_NVMeTCP/185 : FAILURE in 2h 54m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"b23f2bbf8300b4496bea3215def2157bf079e058","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-14 02:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/918805/1/fc : SUCCESS in 0h 38m 51s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/918805/1/iscsi : SUCCESS in 0h 45m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b1ef924b3c5bfbd6cd19df7119a5d726ce17cc07","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-17 15:00:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d035b706137734541dabe317a3e5ea3ce1488222","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-05-20 04:36:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38e8fbe8a9b564bc79ae1c54842336381d354c54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 04:36:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2305a7f21c952023cc57d1f1927737e9242fd8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:38:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c76e56e4a2647a3bb8c150786c5130b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bca601e8a8c4998af7b6342943b8bc7 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c20333fef7949d58b250a57ff11391e : SUCCESS in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ae4e838c6ce4514a620e718c0d10046 : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0f2b09cfbe4786b140d48be67bb22d : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/5033dcf2d05d46cdb46d5c24a8ce29c6 : SUCCESS in 1h 00m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0380883b4a345f2b1020f54c659cd7a : SUCCESS in 1h 20m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2c025b0ea1440aab2bfab3ccd977127 : SUCCESS in 1h 05m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/112894cc560f46ca9388c15ebd9e4eb4 : SUCCESS in 1h 05m 04s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/17af3b40bad9497db18d7eb1b64ee126 : SUCCESS in 2h 19m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe5bd537e63140c891b294431a490b42 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f98b42436562dcc095f3630e836ec3e49f42d0c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:38:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91c9821394e0940261642cc903a5ac53aa74d1eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:42:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/303ff9721ede45c2afc195727095ccac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c42fd95332c4ab8808a2e6c7214432c : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89a0c766927631fd3777d763365070b970f5af4c","revisions":{"89a0c766927631fd3777d763365070b970f5af4c":{"kind":"REWORK","_number":1,"created":"2024-05-09 17:09:54.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/05/918805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/918805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/918805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/918805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/918805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/918805/1"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-04-19 15:55:34.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-05-09 15:35:44.000000000","tz":-240},"subject":"ceph backup: Only warn about striping opts when necessary","message":"ceph backup: Only warn about striping opts when necessary\n\nOnly issue this warning message if the striping options\nwere enabled.\n\nChange-Id: I4116412db30602260049a4129cdfbbc5d5c20c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89a0c766927631fd3777d763365070b970f5af4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89a0c766927631fd3777d763365070b970f5af4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
