)]}'
{"id":"openstack%2Fcinder~918838","triplet_id":"openstack%2Fcinder~master~I3053cc04ecdc46cce55744ece07de016564b431b","project":"openstack/cinder","branch":"master","topic":"bump_sphinx","hashtags":[],"change_id":"I3053cc04ecdc46cce55744ece07de016564b431b","subject":"[DNM] Testing docs bump with new Sphinx","status":"ABANDONED","created":"2024-05-09 19:23:13.000000000","updated":"2024-05-10 02:50:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"31a664069e9558a5e426a7841b9e7edf854e7dec","_number":918838,"virtual_id_number":918838,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-09 21:46:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"reviewer_updates":[{"updated":"2024-05-09 19:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 20:52:08.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-09 20:56:31.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-09 21:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-10 02:50:44.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"}],"messages":[{"id":"931146059a38bbd6b861311e82cb9e2929d7f24c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 19:23:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c8d3f99e84baaf00b7c2d0af77dcc36708e5fad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 19:40:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/101f746bb9a946f387e746815f4947ff\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63e6883425dc4414812de82a73142155 : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/13036f163bed4126b2e1b95fb62a7807 : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"faf86deee0e00ebaf5822a01cd5baf1fade325a5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-09 20:52:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/351/.","accounts_in_message":[],"_revision_number":1},{"id":"53c2750493e137d8d0dc39ab9be54ea465de60b5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-09 20:56:31.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/817bd8d7e2af4ed2a71f6ea654e16e72\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/118800fe4cee429a96952653d2a681f2 : FAILURE in 1h 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eeb9e9779ec0cdf2d1804eef7a2c7d33092a2d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 21:46:34.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/807486b28428431cabe9b4097b3b769c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c63b54dd1aca47319c677526fe66ee12 : SUCCESS in 2m 39s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/08f309d5b3954167ab33505fa9170331 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ca15f9df8c49f486bd9e2cab58efd0 : FAILURE in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1e3c1cea9d94a29a05b8f2b7fe86ec5 : FAILURE in 11m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86042cc41498443b97dcde324f52c5fa : SUCCESS in 12m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c6cc301aec045bb8d775898e82dda08 : FAILURE in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1d7f824556342189e00e3631eac760a : FAILURE in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/72b3f66e90a24b179abd473219f73e68 : SUCCESS in 3m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/786ef1789280442db54826b6902d39d3 : SUCCESS in 1h 08m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/550421b74dec46898b0ff3a42b2d5044 : SUCCESS in 1h 24m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31ecbf5c06e541a399fcc35cb22f2d08 : SUCCESS in 1h 04m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f974d9230b44444ca9bea690a5e67928 : SUCCESS in 14m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0147d87cae3b4396b0d04aeedd8e69be : SUCCESS in 16m 49s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b25168e26c42480c82cc460160454eb3 : SUCCESS in 13m 28s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4248bc405bb64af2a49dd9247d7dcd62 : SUCCESS in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a76fc30f30924d9eafbde4bcfe12933a : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a630ee8afbfe42de870076f2ac4e269d : SUCCESS in 5m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/02a2e51f12aa46309774f43bcc639d05 : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/44972886f8024fec9d1c0bcaa81bfcf6 : SUCCESS in 33m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e84db41a3b547018992f4c71dfdf84f : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/82f4e35df5554a72b6bea9db3bc739f5 : SUCCESS in 2h 02m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c56b0c2b72484228a8d29d72a9469468 : FAILURE in 2h 14m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dcfddc84d2214664afd3724bb652e985 : SUCCESS in 1h 30m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8037618ea4254a5995ca72c67eb0f335 : SUCCESS in 1h 56m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/31e33753768743fcadeed3929035886d : SUCCESS in 45m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ddc519374d64b3588af7a9fc6657a18 : SUCCESS in 1h 20m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df22dc91c69a4e5594d248d981982762 : SUCCESS in 1h 41m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f91870e0e5cf466cab4f109e8c8cc604 : SUCCESS in 52m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1bd55f6fba664e1a80be89051cb72854 : SUCCESS in 1h 30m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b5722c103d7642d0b89d3163d466b375 : SUCCESS in 1h 11m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3182341419484c558659f7fbca0dc78f : SUCCESS in 1h 49m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b96d74448e1e43748459476c728d7e5e : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":1},{"id":"eb9a39db0953676abfd1f645aae108b8ead677aa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 23:08:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"31a664069e9558a5e426a7841b9e7edf854e7dec","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-10 02:50:44.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-38-918838-1/FC/ : FAILURE in 7h 27m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-38-918838-1/iSCSI/ : SUCCESS in 3h 58m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"afe7e97381e90e0a6692b78bcdf83bfb762d9fc2","revisions":{"afe7e97381e90e0a6692b78bcdf83bfb762d9fc2":{"kind":"REWORK","_number":1,"created":"2024-05-09 19:23:13.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/38/918838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/918838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/918838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/918838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/918838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/918838/1"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 19:23:10.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 19:23:10.000000000","tz":-300},"subject":"[DNM] Testing docs bump with new Sphinx","message":"[DNM] Testing docs bump with new Sphinx\n\nSphinx updates can be scary.  This change leverages a constraints bump to see what breaks.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/916565\nChange-Id: I3053cc04ecdc46cce55744ece07de016564b431b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afe7e97381e90e0a6692b78bcdf83bfb762d9fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afe7e97381e90e0a6692b78bcdf83bfb762d9fc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
