)]}'
{"id":"openstack%2Fcinder~918954","triplet_id":"openstack%2Fcinder~master~Ic24bdb4146e4373aa7283cf9c525edf864b8ca95","project":"openstack/cinder","branch":"master","topic":"cleanup-ceph-installer","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-05-08 03:25:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic24bdb4146e4373aa7283cf9c525edf864b8ca95","subject":"DNM: Test dependency on devstack-plugin-ceph changes","status":"ABANDONED","created":"2024-05-10 01:05:45.000000000","updated":"2025-05-08 03:25:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2b16dde0a7d0bb73b217df70b436d0198a89625f","_number":918954,"virtual_id_number":918954,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-09 02:10:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-10 01:16:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-10 02:36:05.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-10 02:38:57.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-10 03:08:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-10 04:04:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-10 04:10:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-05-10 04:40:30.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-10 06:04:14.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-10 06:16:19.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-10 10:12:29.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-05-11 09:55:12.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-05-12 00:16:05.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-13 12:46:09.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-08-08 23:45:25.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-09 00:28:53.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-09 00:28:58.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-09 00:40:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-09 01:12:53.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-09 08:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"752f6fe40d0de1fc9c6b7cdbb6be024d6d68c3cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-10 01:05:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f7be8b2f2e3c499978fdf07e6da062e2819192b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 01:16:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c12d025911e2421bb3105d5afa34c762\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b5662c9c26043f9a10b9f7cc32fb9cf : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6896b8a409e54d1db594def9ccc7c599 : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6f1769af540e943415765f2ecce1a2bb339f420","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-10 02:36:05.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/352/.","accounts_in_message":[],"_revision_number":1},{"id":"491724482329894e2108f2fc1a2cbaf207c42391","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-10 02:38:57.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/497b6d80e1304026a766331eff33fa73\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f69307ee5f1241eca56323360468d043 : FAILURE in 1h 27m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5da16f7f9a952a4ceeaf416cb6105a65da4ee1fd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-10 03:08:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/918954/1/check/DellEMC_Unity_FC/81b769b/DellEMC_Unity_FC/1831 : FAILURE in 1h 49m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/918954/1/check/DellEMC_Unity_iSCSI/492028f/DellEMC_Unity_iSCSI/1859 : SUCCESS in 2h 02m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"3b067854f2b6c04c5fe7fd1cacdb886a155814e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 04:04:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/036a08fa21b94cd18897cb4285b659bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b167dcd9854b5d822d7df09f6994e9 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f26c728a5174ae998aabaf9190e847f : SUCCESS in 10m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8fe0d7ecdf94cc383a974f39c3127c0 : SUCCESS in 9m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9912ccfa56994760b10744c7a4f3879d : FAILURE in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2bf6a0f3c414e51b11839775042ecaa : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/415ae8170a20474691edf7449aeed297 : SUCCESS in 1h 09m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4215f4daf506487792aab1a26bba5df8 : SUCCESS in 1h 39m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1382806b88274920b4cc8bb61c91a4a8 : SUCCESS in 1h 02m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fa2db3037d284bf889a791c2e73fd8af : SUCCESS in 16m 48s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fd9e060e0e2f44f3ba0e956490b705b4 : SUCCESS in 14m 42s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4accedffa3274e4d8cb3eb33f6be863a : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2e0a4a5cc70240629ae50b6df514c25e : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e7bd958596ef47edaa1bdb10b3034a64 : SUCCESS in 6m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/21546dde0baa4a7cbe491654d0916057 : SUCCESS in 7m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a52a4f0344e94c5ea8301407a16572fe : SUCCESS in 49m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38960b745ab546e6a68ed38a817fc7e3 : SUCCESS in 6m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f513376ef10d4cff96cc3717c955eb8f : SUCCESS in 2h 05m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5b5642ec8b2d4dad80e8bed3f034f67a : FAILURE in 2h 36m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1d4a15976f034bcfa251f74f2768cb60 : SUCCESS in 1h 38m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f8b05d68d5474226b72f4813a818b337 : SUCCESS in 1h 59m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3c3fa3b5d8a84a099a24962676a75ce9 : SUCCESS in 29m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5ac5ac6e52d241a8b45760357cb7540a : SUCCESS in 1h 11m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6c098394d5646faba0dff81d0f590c8 : SUCCESS in 1h 03m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d94876a682034c22b9e09fc5cc096dff : SUCCESS in 37m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab874bd2d4a44fbfae4edbfa3bbcdee1 : SUCCESS in 1h 22m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fbec9c22ce59450c9102c70ac123e0ec : SUCCESS in 1h 07m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dda8d7c1785f47e4a78d6ce2a920844d : SUCCESS in 1h 34m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69e0197196a84e11b893f0367d9e92f9 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bb743586c92084b402cb4d2fd25e5a852e69acec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-05-10 04:10:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_05_2024_04_04_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"f9b138852a0bbe2de17d762a627053011d18e6c2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-10 04:40:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/918954/1/upstream-check/cinder-tempest-plugin-ontap-fcp/473cb36/ : RETRY_LIMIT in 3m 44s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/918954/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/22d0aa9/ : RETRY_LIMIT in 3m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/918954/1/upstream-check/cinder-tempest-plugin-ontap-nfs/96077d5/ : RETRY_LIMIT in 3m 32s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/918954/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/5905cf2/ : RETRY_LIMIT in 3m 24s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"67c0afc9270a8f05c67e3ce4621134a99d2ba0ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-10 05:54:08.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/918954/1/check/DellEMC_Unity_FC/e83d11d/DellEMC_Unity_FC/1832 : SUCCESS in 1h 32m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/918954/1/check/DellEMC_Unity_iSCSI/2561075/DellEMC_Unity_iSCSI/1860 : SUCCESS in 1h 49m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"655eb68cf75cf0d8db92708f969ff72f965dd6d2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-10 06:04:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/54/918954/1/check/DellEMC_PowerStore_iSCSI/53e3d8c/DellEMC_PowerStore_iSCSI/1308 : SUCCESS in 1h 54m 41s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/54/918954/1/check/DellEMC_PowerStore_FC/c018ced/DellEMC_PowerStore_FC/1478 : SUCCESS in 1h 03m 55s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/54/918954/1/check/DellEMC_PowerStore_NFS/3c0cbc1/DellEMC_PowerStore_NFS/998 : SUCCESS in 1h 10m 12s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/54/918954/1/check/DellEMC_PowerStore_NVMeTCP/e34002e/DellEMC_PowerStore_NVMeTCP/858 : SUCCESS in 1h 59m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"92604033eb2d95e8df93d69912a9c5d9c5dbc361","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-10 06:16:19.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"d0aea7dd2fb9029387337893396296b7e2ac94f8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-10 10:12:29.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-918954-1/FC/ : FAILURE in 9h 6m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-918954-1/iSCSI/ : SUCCESS in 4h 12m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"dad5204d01d53b6471cf52a1204bfa30f958aa5a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-11 00:06:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2e9d6d487d75721e9e60352f9601bf0c4db83e7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-05-11 09:55:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/918954-1/HPMSA-iSCSI/ : FAILURE in 3:12:44\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/918954-1/HPMSA-FC/ : SUCCESS in 1:38:59","accounts_in_message":[],"_revision_number":1},{"id":"368df961401702c27c34c6afd586f69f1433d1dc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-12 00:16:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/918954/1/20240512042740/NEC-ISCSI : SUCCESS in 4h 18m 24s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/918954/1/20240512042740/NEC-FC : SUCCESS in 4h 48m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8e0c4d4386ced99dbfbd170b1cc96e578c309b7f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-13 12:46:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/918954/1/20240513181303/NEC-V-FC : SUCCESS in 3h 23m 34s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/918954/1/20240513181303/NEC-V-ISCSI : SUCCESS in 3h 33m 03s","accounts_in_message":[],"_revision_number":1},{"id":"db02402352c647a32cda670df26c57ee2906b3d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-08 23:45:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6296119e67f205a9224170e8460119d94e190f29","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-08 23:45:25.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/258/","accounts_in_message":[],"_revision_number":2},{"id":"61f5d36dd0d821f534eaaf6b2fa0ee15b49dbd2d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-08 23:45:30.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/258/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"bac029d2d925c9f2dfdd0a2d89fcec31498b421a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 00:11:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07edb50d91334cdab4e9ffe0a8180cfa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b5909890ef64996ae5a6f3f29f1e5bd : FAILURE in 23m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5c5d53ebe86849128f1e0cb3c5e2679a : SUCCESS in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97cb2435fe6f5f1aa2a48ce3dc827dc2b7bf59d3","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-09 00:28:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/918954 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":2},{"id":"756140870a897793d9042d4264144ff5788003c9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-09 00:28:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/918954 : FAILURE in 41m 17s","accounts_in_message":[],"_revision_number":2},{"id":"6bd044223be168ca9cba057cee88816f86c63082","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-09 00:40:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/918954 : SUCCESS in 44m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8b61873edaae84a0b3df1e0658687e8b41add6bf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-09 01:12:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e4b31f73dd8e4500a3dfff4ba78d6809 : SUCCESS in 1h 25m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d7ae51b2eb5f42b2af4e531cdda226e2 : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0f27228621b4b0aedc0dfedd1a5ea3e3cf03ce64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 02:10:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfe5f51baec140438996b8488dfcef33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ed7913de9a4525ac27278a27d46c64 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca5e9799264d4c79a30e54bccbba5191 : SUCCESS in 11m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47826a556c644832ad4cedcc13744b54 : SUCCESS in 9m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0112c908b16b44c0a3b50f21e30dcb67 : SUCCESS in 16m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a22b30586a04e8c87169d29eb8f03af : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/2aac70c71e2141f098232d6016a6a9e3 : SUCCESS in 1h 03m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b11931943b084af182f1e980d7630e4a : SUCCESS in 1h 32m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/266bf13cf2f14dc9af470e628e2c79b8 : SUCCESS in 1h 01m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f9392fe3dd464a728947f89047e65cd2 : SUCCESS in 17m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3cf7332521f74e4a9dcd36b0b6466bf7 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6968eacaf927441fb47311c63874a98e : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/965897b2c0c545a5b5ce39349a68b862 : SUCCESS in 6m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9ecaa704a0e74f069d80c71f52c4000e : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/613aa6387feb4d6cb16cca651f0833b7 : FAILURE in 44m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fcb1919664904f00ab7efec07872ed3c : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee1f7104287c403283940283f16cefd4 : SUCCESS in 1h 55m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bd08752a900448749146c35db63843cd : SUCCESS in 2h 23m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/870d1508300a4811a44c0d7a04ac915b : SUCCESS in 1h 39m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d241b2438e8e45a7bfb6d0ff8501cb5e : SUCCESS in 1h 52m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/28662dee19af4a9683daf9775afab811 : SUCCESS in 29m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c437099afe24dbcaf2433de3fecb8e3 : SUCCESS in 1h 09m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/45bc2145233745de8dfc7838f9840ce4 : SUCCESS in 1h 41m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/18a3118615f04f0ca1065411aa3c00a8 : SUCCESS in 45m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0835ab905e5c4188b1b2bbc3546fd1c3 : SUCCESS in 1h 41m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8c01242be53948ccbb55c0125d0ee878 : SUCCESS in 1h 24m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/530977ed77f34016baa8fff40dcf379f : SUCCESS in 1h 43m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59702509533c47ce86acb4c180776fb6 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2a0c6b601e95fd0edb6f1f4d93a579f53c5c041c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-09 02:16:23.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_08_2024_02_10_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"d3f06abc3a2813118155c44351ecc23ede87d62c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-09 02:36:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/918954/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/85df1b8/ : RETRY_LIMIT in 3m 34s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/918954/2/upstream-check/cinder-tempest-plugin-ontap-nfs/080b3e1/ : RETRY_LIMIT in 3m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c0fe339c9aa3433cb45d6ddd7a8cdeb7ce7b2375","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-09 04:03:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_PowerStore_iSCSI/de68690/DellEMC_PowerStore_iSCSI/1599 : SUCCESS in 1h 28m 16s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_PowerStore_FC/4ab41a2/DellEMC_PowerStore_FC/1795 : SUCCESS in 1h 01m 30s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_PowerStore_NFS/0338d86/DellEMC_PowerStore_NFS/1292 : FAILURE in 1h 22m 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_PowerStore_NVMeTCP/422694e/DellEMC_PowerStore_NVMeTCP/1151 : SUCCESS in 1h 52m 22s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_PowerStore_NVMeFC/8f917e4/DellEMC_PowerStore_NVMeFC/115 : SUCCESS in 1h 12m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"ae6bb11b1d81684d1beb2363036e53cd37730832","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-09 08:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/918954/2/fc : SUCCESS in 0h 41m 20s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/918954/2/iscsi : SUCCESS in 0h 50m 43s","accounts_in_message":[],"_revision_number":2},{"id":"aafedc1a5efeeb75ebe8994bdca65f02a34747ed","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-09 08:12:52.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_Unity_FC/964e167/DellEMC_Unity_FC/2394 : SUCCESS in 2h 12m 44s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_Unity_iSCSI/3011520/DellEMC_Unity_iSCSI/2399 : SUCCESS in 1h 44m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ff505b87cbc6839b22d6bb2a52e89d4e41f391f6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-09 13:08:02.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_PowerFlex_3.6/844fade/DellEMC_PowerFlex_3.6/1880 : SUCCESS in 2h 08m 26s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_PowerFlex_4.5/cbce33b/DellEMC_PowerFlex_4.5/759 : SUCCESS in 2h 11m 05s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/54/918954/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/f10d6a3/DellEMC_PowerFlex_4.x_NVMeTCP/728 : FAILURE in 13m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"a1a6f5c2e0c675efd2b964d2cd8603ada31f5e83","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-11 07:56:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/918954/2/20240811111004/NEC-ISCSI : SUCCESS in 5h 15m 52s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/918954/2/20240811111004/NEC-FC : SUCCESS in 5h 46m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0ba400324a8ebb11646432a088dafd75844c6640","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-14 02:12:56.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/877/.","accounts_in_message":[],"_revision_number":2},{"id":"b1f48524573df5a354d6a7b1b1e34884e61a2769","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-18 11:08:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/918954/2/20240818161046/NEC-V-FC : SUCCESS in 3h 45m 30s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/918954/2/20240818161046/NEC-V-ISCSI : SUCCESS in 3h 58m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2b16dde0a7d0bb73b217df70b436d0198a89625f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-08 03:25:01.000000000","message":"Abandoned\n\nDevstack plugin ceph change has merged; further testing is not needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5559eae6584815675080172d69e9a9fff3697c88","revisions":{"5a3cfb7a630306cb552aed255deac6acb8ead151":{"kind":"REWORK","_number":1,"created":"2024-05-10 01:05:45.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/54/918954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/918954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/918954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/918954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/918954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/918954/1"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-10 01:05:34.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-10 01:05:34.000000000","tz":-420},"subject":"DNM: Test dependency on devstack-plugin-ceph changes","message":"DNM: Test dependency on devstack-plugin-ceph changes\n\nChange-Id: Ic24bdb4146e4373aa7283cf9c525edf864b8ca95\nDepends-On: I5f75cb829383d7acd536e24c70cc4418d93c13bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a3cfb7a630306cb552aed255deac6acb8ead151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a3cfb7a630306cb552aed255deac6acb8ead151"}]},"branch":"refs/heads/master"},"5559eae6584815675080172d69e9a9fff3697c88":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-08 23:45:14.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/54/918954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/918954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/918954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/918954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/918954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/918954/2"}}},"commit":{"parents":[{"commit":"221a66940ebbba97ecfc4f8830cea50adb3a97a3","subject":"Merge \"Hitachi: Prevent to delete a LDEV assigned to multi objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/221a66940ebbba97ecfc4f8830cea50adb3a97a3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-10 01:05:34.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-08-08 23:45:14.000000000","tz":0},"subject":"DNM: Test dependency on devstack-plugin-ceph changes","message":"DNM: Test dependency on devstack-plugin-ceph changes\n\nChange-Id: Ic24bdb4146e4373aa7283cf9c525edf864b8ca95\nDepends-On: I5f75cb829383d7acd536e24c70cc4418d93c13bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5559eae6584815675080172d69e9a9fff3697c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5559eae6584815675080172d69e9a9fff3697c88"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
