)]}'
{"id":"openstack%2Fcinder~919424","triplet_id":"openstack%2Fcinder~master~Ibd1e9c59701072b6da479b610e7e9475e6d5b7dd","project":"openstack/cinder","branch":"master","attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-09-14 13:10:24.000000000","reason":"Someone else replied on the change"},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2025-10-31 10:55:33.000000000","reason":"\u003cGERRIT_ACCOUNT_37470\u003e replied on the change","reason_account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibd1e9c59701072b6da479b610e7e9475e6d5b7dd","subject":"HPE 3par - Add NVMe FC support","status":"NEW","created":"2024-05-15 09:03:12.000000000","updated":"2025-10-31 10:55:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"740ed1b93e589fa60dff94337c2cc27fba3f5b55","_number":919424,"virtual_id_number":919424,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-17 09:40:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":-1,"date":"2024-07-03 19:47:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-07-03 15:29:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 09:04:36.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 09:08:16.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 09:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 09:27:34.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-15 09:30:00.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-15 09:31:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 11:28:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-15 12:49:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-16 09:00:09.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-05-16 23:21:50.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-05-17 06:08:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-20 14:51:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-20 16:20:45.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-05-21 08:58:40.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-22 02:50:10.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-06-17 08:12:16.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-06-17 09:22:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-17 09:47:07.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-06-28 03:22:35.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-06-29 00:56:49.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-07-03 15:29:07.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2024-07-03 19:47:24.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2025-10-31 10:55:33.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"REVIEWER"}],"messages":[{"id":"333acf702613f4b816c2e7a030c5019d75e2d563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-15 09:03:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"793c1de63daa2f2e9cc6cae104a34afe0ec48ef1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-15 09:04:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/24/919424/1/check/DellEMC_PowerFlex_3.6/2a7e46e/DellEMC_PowerFlex_3.6/1351 : FAILURE in 27s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/24/919424/1/check/DellEMC_PowerFlex_4.5/147ddf6/DellEMC_PowerFlex_4.5/276 : FAILURE in 57s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/24/919424/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/d1e0423/DellEMC_PowerFlex_4.x_NVMeTCP/222 : FAILURE in 1m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"49eda9c59ca025cc265ccbfc2d53522a3087798c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-05-15 09:08:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/24/919424/1/check/fujitsu-eternus-dx-iscsi/7cfbba2 : FAILURE in 3m 05s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/24/919424/1/check/fujitsu-eternus-dx-fc/c4c11a3 : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4803f88aeb9b8b39533b383ba3ad424c953baf86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 09:16:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c31fba8c3d641f7b34f5d293e83ed81\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7824f8e3515f454dbdf2e02c768fae57 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/83632c72492342a8a85868dae86eead4 : SUCCESS in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c85185390bcfdfd629507f8df46834be907881eb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-15 09:27:34.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3fa853cc84654f89b25c1cfa69338f2b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a5a8d804a53045e48be6976f4c386cda : POST_FAILURE in 20m 43s","accounts_in_message":[],"_revision_number":1},{"id":"038836cecdd510644c0d38c8e1341016b32030ad","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-15 09:30:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/919424 : FAILURE in 25m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0e9a1837ac49a428facf405e773e33e5bf544290","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-15 09:31:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/919424 : FAILURE in 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e342f97f13e4325a2aab15febaf22869dd378714","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-15 11:28:19.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/388/.","accounts_in_message":[],"_revision_number":1},{"id":"c8ca429b263609a8fe85e6250354469759fd5599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 12:49:57.000000000","message":"Patch Set 1: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78828adbe7494ac3bfe868280d302ece\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c970931ccf40e9a6d5f7c950957855 : FAILURE in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7cf267d13674b838ecd219188c02d4e : SUCCESS in 10m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45140d151ed24fcb9b5b1b5a3e29c9d4 : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05a2daa0d1234991b875d6c6e794a4da : FAILURE in 10m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28edd4ef29b347c79d6d570e70c71cea : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/bdaee451d7d84ac8bc0ca893b4899a0f : SUCCESS in 55m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/893ed618692043549da34dfb554659ac : SUCCESS in 1h 31m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cacb55d5b76549fe8232989cc136cfcd : SUCCESS in 50m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4dcabf4848264ab5a62efca57b137b1d : SUCCESS in 15m 27s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d48eaddb640c41578971dc66bbe2f6d3 : SUCCESS in 15m 29s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3b32e93fbca04327a1226b6bcc3a9ff7 : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a486be8332b14537ab96cee10a0627b2 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/08c41315d9504daeb916b0edf896eede : SUCCESS in 6m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fdd86acf8c724bb68d071a7b27a08118 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c3044c1f032248a1ab227d28cb715a53 : FAILURE in 34m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54783333d24b4c90a9f23d62672f23fd : FAILURE in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc114885ab5a4e77aec2bdb3349a80fe : SUCCESS in 1h 49m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5d177e81ef444b00b0e7fd42e31868ff : FAILURE in 1h 49m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4eed6a3a8d2341d4bde50e44d43e7f38 : FAILURE in 1h 47m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3e2bacb1cea748cdac757f7324b5f283 : TIMED_OUT in 3h 10m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/50d091efea6e4a32bf87d34a8a4824ae : SUCCESS in 40m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/30f8a8c28f2e47f6bdd63e031b5d36fc : SUCCESS in 1h 19m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8dda890e315424aa188360b1309c702 : SUCCESS in 1h 17m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5a09a9fe1ba04a249fb6821b171a2e2c : SUCCESS in 47m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6544c7b8d28a4f1d973431c83d22bde0 : SUCCESS in 1h 18m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/addb5604b18c441c8ff53002d26cae60 : SUCCESS in 1h 32m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5199a4e6bf934617b6e047c82f35713b : SUCCESS in 1h 24m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0c773ab1e7142f097e3cdf2ef9afba3 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":1},{"id":"54bb9153e422df7bd2f326e295127cc304b4d8d1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-16 09:00:09.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/24/919424/1/fc : SUCCESS in 0h 39m 29s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/24/919424/1/iscsi : SUCCESS in 0h 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"bf8ae166e7fa82560efd43010631b36460098fe0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-16 23:21:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/1/20240517045120/NEC-V-FC : SUCCESS in 3h 15m 59s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/1/20240517045120/NEC-V-ISCSI : SUCCESS in 3h 30m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9dfc05c8f09b5852e33bca7c3b2e6ed0ea56d76a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-17 06:08:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/1/20240517080412/NEC-FC : SUCCESS in 4h 51m 37s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/1/20240517080412/NEC-ISCSI : SUCCESS in 7h 04m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0a13a1737488549b86c2663b4e80406461170312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-20 07:15:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee12fc0eea8ac333781450af5e02c56e30340da2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 07:30:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04cc62ea85f040da987b13aeede9fc60\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb5c1a04e2894fcf8acfca62f8106729 : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44a227990a284ed9b76f08f0dec98ce5 : SUCCESS in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bafe1d4c63ad1d88bfad4d7cc9adda42a22a9f8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 07:39:24.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8592a2cf36c24c6d979798e9eba9dca7\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f951658bb107488aa7acfa7efa6c642f : POST_FAILURE in 20m 17s","accounts_in_message":[],"_revision_number":2},{"id":"d50b080dd3d6444f94537fcd24ec651bf2baec5b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 07:40:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/919424 : FAILURE in 23m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2561b6bfb8a1c91ad6eff7f143100b38fa2ae731","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 08:02:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/919424 : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":2},{"id":"072dd3fa187d799b99eb8d1ffcfa64df80f5ecfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-20 09:44:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fae32368a3d37f463d4298ebcb77ccff53c05ece","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:00:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0a2af607c8c4f7584b65af2f7fe82b5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46faf0e01a534d4284f8b23f5e45c1ed : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8394afeb8b5548b8b135807aed5bd1fe : SUCCESS in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c31d457cbe15c6b44cdfb3f3eb6873110061f30e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 10:09:05.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/919424 : FAILURE in 22m 42s","accounts_in_message":[],"_revision_number":3},{"id":"23a0ff52976f15a9e817ad33bb66b2f7d9a32790","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 10:11:30.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/26c7fa9cf8f74a2082aee5d1a0514c97\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3b6a91b40b9e4faf8aab0117e2d545a1 : POST_FAILURE in 20m 43s","accounts_in_message":[],"_revision_number":3},{"id":"21915006a6def07e675c9e068d726c7bf21d0a0f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 10:31:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/919424 : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":3},{"id":"6a4562309a5425ba194c7bec5be6d5d8e4d7215d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-20 11:45:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/424/.","accounts_in_message":[],"_revision_number":3},{"id":"58bb2ff8b587c5b2cf74cf6097038660541c5312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 12:18:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca51478b5efc4fb3a1a7adce52d18e78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a24caa56cd74a5685e917974fe5b47d : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae7a1756f9394f5cbd23d64e0eb2c833 : SUCCESS in 11m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ec95f37f8b048bdb5979c57df2b6728 : SUCCESS in 13m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28011685c4e541728e349cb34124a6f5 : FAILURE in 10m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a5e8dea12e426e9dfed59dbdf4f231 : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/1fbd9e750657409daf29894ad8473870 : SUCCESS in 46m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df243b9fe7dc444faebdbfcaed1ba0ce : SUCCESS in 1h 30m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43e34b14787949ed91578ac6f027c030 : SUCCESS in 1h 03m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f4dc006ddc50434d946076d3dcff69f9 : SUCCESS in 15m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6eb58eb108c645c9b25a2fca98e2c1d3 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/efe6eec0b4fb473ca14d49d5fc610899 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78606509c4ae46dd932dd7a1e4182c0d : SUCCESS in 6m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8e0cc582c11d479fad45389266e1dbef : SUCCESS in 7m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3b14734739a34c56b6c8a2c19c170238 : FAILURE in 35m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1dbd661084b94c83b0d1025062fff460 : FAILURE in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32e6a17d5ce9451d8f4ca2a1fa5c122a : SUCCESS in 1h 51m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/67fb95580dac4dc4aceb126e3494c03e : FAILURE in 2h 27m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/01c18f73d8e740e68b02a4ec387273ce : FAILURE in 2h 05m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9dcda0ffcb8849158c2bd31728be74db : SUCCESS in 1h 43m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/508b76bf00f94e3abcfb8c79aeeadb8e : SUCCESS in 28m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1a3ffb3e2e2a49c69747aab55e316a56 : SUCCESS in 1h 10m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cf71b3a07ea84e519473e5cc8eac6726 : FAILURE in 1h 34m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/67f9f2f260f54d88886cf2972221c7bc : FAILURE in 36m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/93f434fab1444a27a2736f7aa2dabc3d : SUCCESS in 1h 19m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/404fac76672d4cb580c5c17bbdfdbb71 : SUCCESS in 1h 28m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0915f7b6bec946a9a4af60f6b9c85225 : SUCCESS in 1h 39m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a032217c766496aad1a9fb96cb5fb09 : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":3},{"id":"b10c5afc952fd2ea03c10f19f1260db10af54c30","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-20 13:00:10.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/24/919424/3/fc : SUCCESS in 0h 40m 24s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/24/919424/3/iscsi : SUCCESS in 0h 50m 48s","accounts_in_message":[],"_revision_number":3},{"id":"40c824c0cbecc7c0028db177d518c59727a4cac8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-20 14:51:49.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/24/919424/3/check/DellEMC_Unity_FC/fc0fdc5/DellEMC_Unity_FC/1901 : FAILURE in 1h 27m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/24/919424/3/check/DellEMC_Unity_iSCSI/27d7d48/DellEMC_Unity_iSCSI/1929 : SUCCESS in 2h 05m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"631335c0af030181006d187e6f09bf8a13446ee7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-20 16:20:45.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-24-919424-2/FC/ : FAILURE in 9h 4m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-24-919424-2/iSCSI/ : FAILURE in 5h 27m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"65c80d11b0dd3a1b22f53b4ba6bde5ba12a190ab","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-20 19:01:52.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-24-919424-3/FC/ : FAILURE in 9h 16m 33s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-24-919424-3/iSCSI/ : SUCCESS in 5h 8m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"fb29f3d455c247b097557a1a0d4d6993f11e3aa8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-20 20:48:49.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/24/919424/3/check/DellEMC_PowerFlex_3.6/3bc0834/DellEMC_PowerFlex_3.6/1400 : SUCCESS in 1h 54m 28s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/24/919424/3/check/DellEMC_PowerFlex_4.5/9fdba61/DellEMC_PowerFlex_4.5/318 : SUCCESS in 1h 52m 33s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/24/919424/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/54ced12/DellEMC_PowerFlex_4.x_NVMeTCP/263 : FAILURE in 2h 55m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"9d0d0418400e31d4e6d9f64bf6f4a0601f34dfa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-21 06:14:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b3d3ad4836866c224bf0aaa67f40912f18995f2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 06:37:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54dd7a0f3b34c399b95d27eee706d46\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9959bc9368e244c298624a618948b0a8 : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/201d4bd1f0d742dc9ef9b042e366f751 : SUCCESS in 19m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ba0d18852b88adfcf1855fd119a65b162a54d796","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-21 07:50:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/3/20240521115740/NEC-ISCSI : SUCCESS in 4h 23m 04s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/3/20240521115740/NEC-FC : SUCCESS in 4h 52m 34s","accounts_in_message":[],"_revision_number":3},{"id":"44eafe715c117a01c87c58746680f85d800366cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:38:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f57c2f91fa094392a6d3eb551761d3dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba555bef0a334da89b9b61ac04f39c59 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8486c447f67a49129f11fc17c3c2ad63 : SUCCESS in 11m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c555dbf54b24d60ac11f263ddc7d36e : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b54e55899974df5b163c364b1fe8ef9 : FAILURE in 9m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc1f0ddeefc427288e26605cd6292a4 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/434d5cb529c945c991a02541b784077f : SUCCESS in 59m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9af00bd404443fda22acdcf64c4cb0d : SUCCESS in 1h 29m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/805019e912c0455e94941b6cf230a9aa : SUCCESS in 1h 03m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e22e35464b049daa388f0f970fc91d7 : SUCCESS in 11m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6bf388a4a8584675bd21069e028ea24b : SUCCESS in 14m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b9730b945b004594ad807e15cc7f205b : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fee86ab48e5e49c3b3a6670ded54947f : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31e7cce15ca94518868c34e9779210dc : SUCCESS in 6m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d9e512b05d204960a990918c6cb11e4c : SUCCESS in 9m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e07a033959ff468692484a7b19d5b783 : FAILURE in 42m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cecba0e89f924f98932553f6f905d4a7 : FAILURE in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b1932fc823d4b88a88346057f0fdf57 : SUCCESS in 1h 55m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/122840503f824470bd392352a5c31f56 : FAILURE in 2h 10m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3c480631b36247dfa7caae2eebc0b028 : SUCCESS in 1h 47m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2c39806786fd4afe96ec909bd2c75de1 : SUCCESS in 1h 51m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4bb2b1e90abe4ee8b573ae37cd9e94a8 : SUCCESS in 36m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8c8512bfa44b4bf8b125cc7a6fd3642f : SUCCESS in 1h 19m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/674be4828d534b99bf9f1d634ed8aa5c : SUCCESS in 1h 21m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38a6e80a49ad4cfc9b4754a202706b5c : SUCCESS in 33m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/368b4ddc7814450591e44fe396bb9bf3 : SUCCESS in 1h 21m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8dcbd2f3c14d4bfa921fec3825ed7e27 : SUCCESS in 1h 03m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/597b11a4bbf34a9abaa638b10009d624 : SUCCESS in 1h 31m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77a8562b05d64ea3be163204eea87e55 : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":4},{"id":"dc117c3b29dd32fe99b52d094276a0c22001d474","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-21 08:58:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/919424/4/upstream-check/cinder-tempest-plugin-ontap-fcp/db4a214/ : RETRY_LIMIT in 3m 51s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/919424/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/08f9a60/ : RETRY_LIMIT in 3m 29s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/919424/4/upstream-check/cinder-tempest-plugin-ontap-nfs/7943d98/ : RETRY_LIMIT in 3m 51s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/919424/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/b0fe422/ : RETRY_LIMIT in 3m 45s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"dbc58aff0d16cbae9055eb80e013b1187124ec8c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 09:07:14.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/adedb126fd53494aaa5482cf3417a284\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/70553f640dda47ffa5eca6419fa54fd2 : POST_FAILURE in 20m 51s","accounts_in_message":[],"_revision_number":4},{"id":"5fdd369ed1acec860916e221493d0f9d1a589053","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-21 09:13:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/919424 : FAILURE in 23m 20s","accounts_in_message":[],"_revision_number":4},{"id":"332266184e7f7ad6f391d6de269ce3b7be7c28f6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-21 10:04:10.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/3/20240521111831/NEC-V-ISCSI : SUCCESS in 3h 48m 51s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/3/20240521111831/NEC-V-FC : FAILURE in 7h 45m 35s","accounts_in_message":[],"_revision_number":3},{"id":"38c05d51921ffa38787af34d2dfa07ceb8dc841e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-21 12:09:55.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/919424 : SUCCESS in 44m 00s","accounts_in_message":[],"_revision_number":4},{"id":"101072e355fb61014e231121a960e6bacf659505","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-21 22:00:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_PowerFlex_3.6/18f0a93/DellEMC_PowerFlex_3.6/1413 : SUCCESS in 1h 42m 38s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_PowerFlex_4.5/c726460/DellEMC_PowerFlex_4.5/329 : FAILURE in 1h 54m 20s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/a0035c7/DellEMC_PowerFlex_4.x_NVMeTCP/270 : FAILURE in 2h 55m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"9960cdb734ee3689613cde317b8617d883e9ebeb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-22 02:50:10.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_PowerStore_iSCSI/2cdf805/DellEMC_PowerStore_iSCSI/1363 : SUCCESS in 1h 11m 48s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_PowerStore_FC/234372d/DellEMC_PowerStore_FC/1546 : SUCCESS in 1h 02m 02s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_PowerStore_NFS/88bbc67/DellEMC_PowerStore_NFS/1057 : SUCCESS in 59m 52s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_PowerStore_NVMeTCP/c286eca/DellEMC_PowerStore_NVMeTCP/911 : SUCCESS in 1h 41m 22s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"ac69980766416b840e84cc727cdfba4c7ef8cb61","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-22 12:27:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/457/.","accounts_in_message":[],"_revision_number":4},{"id":"2976209039aa2957185e7ce049eeeae79a103f49","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-23 20:56:10.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_Unity_FC/f9b87d6/DellEMC_Unity_FC/1937 : FAILURE in 2h 06m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/24/919424/4/check/DellEMC_Unity_iSCSI/328637d/DellEMC_Unity_iSCSI/1965 : FAILURE in 2h 09m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e6b07e962946749fe2c7c63de9b2ad47f39808a3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-28 12:27:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/4/20240528133355/NEC-FC : SUCCESS in 5h 28m 07s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/4/20240528133355/NEC-ISCSI : SUCCESS in 7h 53m 22s","accounts_in_message":[],"_revision_number":4},{"id":"41294c0e611e638d683c6200035257329ed53047","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-30 02:03:23.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/4/20240529235821/NEC-V-FC : FAILURE in 7h 46m 14s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/919424/4/20240529235821/NEC-V-ISCSI : FAILURE in 11h 04m 57s","accounts_in_message":[],"_revision_number":4},{"id":"5c2e6969606550fd5c0ffbc04969054300435093","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-31 00:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/24/919424/4/fc : SUCCESS in 0h 45m 57s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/24/919424/4/iscsi : SUCCESS in 0h 54m 52s","accounts_in_message":[],"_revision_number":4},{"id":"c4682096039f32b03459c2ecedb8f847b2c9962a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-17 07:06:12.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c489dee5b7efaa19034a99922c58dfceef5c0466","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-17 07:07:24.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"2c57a21ffd3b490fabb4d833def595ca747433f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-17 07:10:42.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"b48275a8ab17a5ecf0b0879e6cd816e158bc1dcb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-17 07:11:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"7cc8b42b4d19d81dbf194ad2c77aa01fe1bd3337","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 07:33:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83cb5ec8962a4b1e8160a079dbfcd0d2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96a25f221efb47f1a1374949efa4fe2a : SUCCESS in 21m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b9947d280a1412cb9e0e3239fb0ed5b : SUCCESS in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ea1391fcf819ddc0b59af2ff45b89a22a8108e21","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-17 07:36:50.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3d49244c3cdb484aa7da3f95845dcab8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/27cf534ec432465f98615c1eb08c929f : POST_FAILURE in 20m 18s","accounts_in_message":[],"_revision_number":6},{"id":"2352431aa3c438d481ca738fc456dcc346770775","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-17 08:12:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/919424 : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":6},{"id":"c580bbc1f418e5c20c23163206788f46ffcc0e2e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-17 08:30:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/919424 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":6},{"id":"71330f6ee86fbec6420902c5df9275d29033aea0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-17 08:42:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/130/.","accounts_in_message":[],"_revision_number":6},{"id":"8a379df5729316c735f7c6d678348b10db57c9c8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-17 08:52:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/616/.","accounts_in_message":[],"_revision_number":6},{"id":"1faa5410b99bcbc846dcc2754ed44aa6cab15b30","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-17 09:22:08.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e1167b2b392e4f3f9d2da326acf699e1 : FAILURE in 1h 11m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/acbf42b9ad5b4af4ac8c1245d353f92d : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":6},{"id":"2300d120f066b801d0a1834370c823142aea5a0c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-17 09:39:38.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerFlex_3.6/8fe1785/DellEMC_PowerFlex_3.6/1592 : SUCCESS in 2h 11m 47s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerFlex_4.5/b25e052/DellEMC_PowerFlex_4.5/487 : SUCCESS in 1h 32m 29s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/1570c15/DellEMC_PowerFlex_4.x_NVMeTCP/437 : FAILURE in 11m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"cb0ddb65e3bde4561fe30cae29b6ea8f9eea6fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 09:40:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0879cc6474934ad998aaa5d367b398a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64138e8ebf724372aaa8f6b1a6427fae : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ec4304c64bd410baba75ff0b6654688 : SUCCESS in 11m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d857163aac742d48143b0a92e14bb82 : SUCCESS in 10m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5db0cf780a954735a5232ac652b94408 : SUCCESS in 15m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da669b47ee874b27a546bc58c4277d9b : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/49f2824d44b946c1a7afe60bce26c957 : SUCCESS in 1h 01m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21f9aedd99a54cef958b63abf01a0fbc : SUCCESS in 1h 28m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc9b846731b04b3dba99bd1e71d1dfc9 : SUCCESS in 58m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b163e9d311da43afa5f5503c8cb311f9 : SUCCESS in 17m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/92192f55aff040458d2b91e80655586e : SUCCESS in 16m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/db52f69bebd64ec3b31eac17ac500507 : SUCCESS in 4m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb7fa8eba1ad44f5937344648914efab : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/902aba3820bf433b957c825d1f7a0262 : SUCCESS in 5m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5edd502dc66d437ca7ac244e8222da4b : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4c3293be77934542a197e653477efb98 : FAILURE in 43m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e537d2887c94a01905fe8640c92df74 : FAILURE in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/39351a30711346c8b7a5deaa14a26e46 : SUCCESS in 2h 08m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1bc72a920dca4f9bb387f42c4064aff7 : FAILURE in 2h 22m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6c6eb8ac1d29413b9d628e71bffef472 : SUCCESS in 1h 45m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/90c993c5e55a405dbac2192d42cdc079 : SUCCESS in 1h 56m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c4a852f84ced44a3a0d547f7531b274c : SUCCESS in 37m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b0cae19a8d07434684eb865c35e66f81 : SUCCESS in 1h 06m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a8cba01d48fe4cd9baa993808d275cd4 : SUCCESS in 1h 34m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ff6977d8bea44a9a79ce7fdec814b96 : SUCCESS in 37m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/be7609ad33544adf94633e4e17d95874 : SUCCESS in 1h 12m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b3890e19470a4fd190689808582dae75 : SUCCESS in 1h 29m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c684124c2c624d63b79aca4df29b5686 : SUCCESS in 1h 28m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73137de64bb6485da0a4f27fa8ac3b0c : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":6},{"id":"d678f2221ad15ca94cf0bef3c5bf451f59aba840","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-06-17 09:47:07.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_06_2024_09_40_58\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"735ab209e8380a7525f661fd1d7ddcf29a1dfa49","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-17 10:22:43.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-24-919424-5/FC/ : SUCCESS in 2h 46m 19s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-24-919424-5/iSCSI/ : SUCCESS in 2h 14m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"b65fdb32c52eb8036cde8dd569ce5af12035e243","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-17 11:33:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/919424/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/4231986/ : SUCCESS in 46m 16s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/919424/6/upstream-check/cinder-tempest-plugin-ontap-nfs/7603777/ : SUCCESS in 1h 35m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"bf60a5cdabd06bc4b0382389bf1ea22c4a459c7f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-17 12:26:59.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerStore_iSCSI/234da6a/DellEMC_PowerStore_iSCSI/1450 : SUCCESS in 1h 22m 58s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerStore_FC/0cff5e5/DellEMC_PowerStore_FC/1633 : SUCCESS in 58m 41s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerStore_NFS/410a67a/DellEMC_PowerStore_NFS/1144 : SUCCESS in 1h 12m 11s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerStore_NVMeTCP/f295c52/DellEMC_PowerStore_NVMeTCP/997 : FAILURE in 2h 32m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"fc1a48d216340f911d747134da266def50d3a88c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-17 16:09:33.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-24-919424-6/FC/ : FAILURE in 8h 3m 1s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-24-919424-6/iSCSI/ : SUCCESS in 2h 6m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"e006adc5a6e7ebce33f2f2be29237ca26a8cae17","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-17 17:30:10.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerFlex_3.6/7162734/DellEMC_PowerFlex_3.6/1596 : SUCCESS in 2h 01m 14s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerFlex_4.5/3eb8d43/DellEMC_PowerFlex_4.5/491 : SUCCESS in 1h 44m 12s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/95cca8c/DellEMC_PowerFlex_4.x_NVMeTCP/441 : FAILURE in 11m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"25371be9e785ebe44d4ee2897291cd1c4eb02659","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-18 05:29:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8aa18d46d6ae3798f0bb04807c60cd30b2c56ccd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-18 14:17:29.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_Unity_FC/0cee87b/DellEMC_Unity_FC/2101 : SUCCESS in 1h 28m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/24/919424/6/check/DellEMC_Unity_iSCSI/6a5f885/DellEMC_Unity_iSCSI/2121 : SUCCESS in 2h 09m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"76f0953631f9ae629528f4390a21eacbaa617c28","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-06-28 03:22:35.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/919424-6/HPMSA-iSCSI/ : SUCCESS in 3:13:06\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/919424-6/HPMSA-FC/ : SUCCESS in 2:28:23","accounts_in_message":[],"_revision_number":6},{"id":"772b6042e5adeaa7ba9671338f5039fa667d2e31","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-06-29 00:56:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/919424-6/Seagate-iSCSI/ : SUCCESS in 2:17:50\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/919424-6/Seagate-FC/ : SUCCESS in 2:13:05","accounts_in_message":[],"_revision_number":6},{"id":"f5664f695dcc93fb692d3e2c34620f2576592a44","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2024-07-03 15:29:07.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f13c23b5f4df0794e79d466e446c091235eb2abc","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-07-03 19:47:24.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"54ff8f0ba9db72fe165b77ebc470463c319a14be","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-04 06:00:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f0da617b559871d6269060891b297bcfe8d2eeb2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-14 13:10:24.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/24/919424/6/20240914143044/NEC-V-FC : SUCCESS in 3h 54m 39s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/24/919424/6/20240914143044/NEC-V-ISCSI : FAILURE in 7h 39m 36s","accounts_in_message":[],"_revision_number":6},{"id":"beacf3c4c7ee9acb78df23b667643d0b0347b5a3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-15 06:55:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/24/919424/6/20240915101350/NEC-ISCSI : SUCCESS in 5h 14m 24s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/24/919424/6/20240915101350/NEC-FC : SUCCESS in 5h 41m 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cccbe93133ee228715bc8f0fb0f76011d1b4e8fe","revisions":{"d514898a4bbae363762cbff44fc1d99b7fdfa276":{"kind":"REWORK","_number":1,"created":"2024-05-15 09:03:12.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/24/919424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/919424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/919424/1"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-15 09:05:29.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-15 09:06:55.000000000","tz":0},"subject":"[WIP] - HPE 3par - Add NVMe FC support","message":"[WIP] - HPE 3par - Add NVMe FC support\n\nWork in progress.\nThis patch adds support for NVMe FC protocol.\n\nChange-Id: Ibd1e9c59701072b6da479b610e7e9475e6d5b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d514898a4bbae363762cbff44fc1d99b7fdfa276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d514898a4bbae363762cbff44fc1d99b7fdfa276"}]},"branch":"refs/heads/master"},"2a5b10556d38ed057163e2f551e21cd3b7b9d633":{"kind":"REWORK","_number":2,"created":"2024-05-20 07:15:57.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/24/919424/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/919424/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/919424/2"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-15 09:05:29.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-20 07:19:28.000000000","tz":0},"subject":"[WIP] - HPE 3par - Add NVMe FC support","message":"[WIP] - HPE 3par - Add NVMe FC support\n\nWork in progress.\nThis patch adds support for NVMe FC protocol.\n\nChange-Id: Ibd1e9c59701072b6da479b610e7e9475e6d5b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a5b10556d38ed057163e2f551e21cd3b7b9d633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a5b10556d38ed057163e2f551e21cd3b7b9d633"}]},"branch":"refs/heads/master"},"1206894944dba41ec1d13bec393ebae763b5c723":{"kind":"REWORK","_number":3,"created":"2024-05-20 09:44:59.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/24/919424/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/919424/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/919424/3"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-15 09:05:29.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-20 09:48:34.000000000","tz":0},"subject":"[WIP] - HPE 3par - Add NVMe FC support","message":"[WIP] - HPE 3par - Add NVMe FC support\n\nWork in progress.\nThis patch adds support for NVMe FC protocol.\n\nChange-Id: Ibd1e9c59701072b6da479b610e7e9475e6d5b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1206894944dba41ec1d13bec393ebae763b5c723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1206894944dba41ec1d13bec393ebae763b5c723"}]},"branch":"refs/heads/master"},"34947edca0c3cf87a5957863652b9c20d9fc853c":{"kind":"REWORK","_number":4,"created":"2024-05-21 06:14:22.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/24/919424/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/919424/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/919424/4"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-15 09:05:29.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-21 06:16:06.000000000","tz":0},"subject":"HPE 3par - Add NVMe FC support","message":"HPE 3par - Add NVMe FC support\n\nThis patch adds support for NVMe FC protocol.\n\nThe NVMe ports are used over FC ports as shown in output below.\nWhile creating host entry on array, NQN of host is used.\nFor attach volume to host, the FC ports (and its wwn) are used.\n\ns042N cli% showport -vpi\nN:S:P   Mode     State --Node_WWN/IP--- -Port_WWN/HW_Addr- Type Protocol VPI\n0:3:1 target     ready 2FF70002AC02B114   20310002AC02B114 host       FC   0\n0:3:1 target     ready 2FF70102AC02B114   20310102ADF2B114 free     NVMe  31\n1:3:1 target     ready 2FF70002AC02B114   21310002AC02B114 host       FC   0\n1:3:1 target     ready 2FF70102AC02B114   21310102ADF2B114 free     NVMe  31\n\nChange-Id: Ibd1e9c59701072b6da479b610e7e9475e6d5b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34947edca0c3cf87a5957863652b9c20d9fc853c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34947edca0c3cf87a5957863652b9c20d9fc853c"}]},"branch":"refs/heads/master"},"475b03997ac061c895dbff35afd92e36ce1f557c":{"kind":"REWORK","_number":5,"created":"2024-06-17 07:06:12.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/24/919424/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/919424/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/919424/5"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-15 09:05:29.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-06-17 07:09:05.000000000","tz":0},"subject":"HPE 3par - Add NVMe FC support","message":"HPE 3par - Add NVMe FC support\n\nThis patch adds support for NVMe FC protocol.\n\nThe NVMe ports are used over FC ports as shown in output below.\nWhile creating host entry on array, NQN of host is used.\nFor attach volume to host, the FC ports (and its wwn) are used.\n\ns042N cli% showport -vpi\nN:S:P   Mode     State --Node_WWN/IP--- -Port_WWN/HW_Addr- Type Protocol VPI\n0:3:1 target     ready 2FF70002AC02B114   20310002AC02B114 host       FC   0\n0:3:1 target     ready 2FF70102AC02B114   20310102ADF2B114 free     NVMe  31\n1:3:1 target     ready 2FF70002AC02B114   21310002AC02B114 host       FC   0\n1:3:1 target     ready 2FF70102AC02B114   21310102ADF2B114 free     NVMe  31\n\nImplements: https://blueprints.launchpad.net/cinder/+spec/hpe-3par-nvme-fc-support\nChange-Id: Ibd1e9c59701072b6da479b610e7e9475e6d5b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/475b03997ac061c895dbff35afd92e36ce1f557c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/475b03997ac061c895dbff35afd92e36ce1f557c"}]},"branch":"refs/heads/master"},"cccbe93133ee228715bc8f0fb0f76011d1b4e8fe":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-06-17 07:10:42.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/24/919424/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/919424/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/919424/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/919424/6"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-15 09:05:29.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-06-17 07:13:30.000000000","tz":0},"subject":"HPE 3par - Add NVMe FC support","message":"HPE 3par - Add NVMe FC support\n\nThis patch adds support for NVMe FC protocol.\n\nThe NVMe ports are used over FC ports as shown in output below.\nWhile creating host entry on array, NQN of host is used.\nFor attach volume to host, the FC ports (and its wwn) are used.\n\ns042N cli% showport -vpi\nN:S:P   Mode     State --Node_WWN/IP--- -Port_WWN/HW_Addr- Type Protocol VPI\n0:3:1 target     ready 2FF70002AC02B114   20310002AC02B114 host       FC   0\n0:3:1 target     ready 2FF70102AC02B114   20310102ADF2B114 free     NVMe  31\n1:3:1 target     ready 2FF70002AC02B114   21310002AC02B114 host       FC   0\n1:3:1 target     ready 2FF70102AC02B114   21310102ADF2B114 free     NVMe  31\n\nImplements: blueprint hpe-3par-nvme-fc-support\nChange-Id: Ibd1e9c59701072b6da479b610e7e9475e6d5b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cccbe93133ee228715bc8f0fb0f76011d1b4e8fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cccbe93133ee228715bc8f0fb0f76011d1b4e8fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
