)]}'
{"id":"openstack%2Fcinder~919848","triplet_id":"openstack%2Fcinder~master~Idf93686a13cfc40923bc142cad058c97fecf5177","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-05-16 21:10:42.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-05-16 19:46:55.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Idf93686a13cfc40923bc142cad058c97fecf5177","subject":"Fix fast8 tox env","status":"MERGED","created":"2024-05-16 14:25:57.000000000","updated":"2024-05-16 22:58:38.000000000","submitted":"2024-05-16 21:10:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919848","meta_rev_id":"f217d112fcc9ed30fcce56d768217f97e8231e48","_number":919848,"virtual_id_number":919848,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-16 21:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2024-05-16 18:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-16 19:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-16 19:46:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 14:26:17.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-05-16 14:37:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-16 14:41:10.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-16 15:23:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-16 16:12:50.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-16 16:25:48.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-16 18:24:18.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-16 18:41:00.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-05-16 19:13:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-16 22:58:38.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"}],"messages":[{"id":"e22b0ff468d7896619767c93776292bc0709bcf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-16 14:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5157df800b5ecf79e7b019bbe12683d6fa2fbc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:37:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1fa18cb203140039f759e6f50622e03\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9310f8b60c448608f3a21a731669714 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47c8b3cbd4e44e0a80ebcb7fea38a782 : SUCCESS in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f819af6406ab0f4a139b591c697751edf64d24e4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-16 14:41:10.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/041bfe7d40254860a426eb5eedd37f20\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3ffd2d8903054457ba657e20c49f48ba : POST_FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"03819dc9ea268c88c31d7d85d5151c015fc53d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 15:23:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d98a5628c0d4967b715c4ae36e58eb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2f37a0906e4a65a9875ffda1d4c31e : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/780ff884d4474206b87e7d0ebe589a18 : SUCCESS in 11m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06e7ed7fda78436a982ce3ad917e987b : SUCCESS in 10m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/443476e71a414990b3540740bde6501c : FAILURE in 9m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce8f2c82e7646248f95dc6574cc40f0 : SUCCESS in 13m 53s\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/eabd0be3bf44441eac887f1ec4b0b9f1 : SUCCESS in 13m 14s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/39f56b7e6ffc451f9a36de767388eb11 : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3dba94533250464b8aee689f1264a8b6 : SUCCESS in 5m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/01d588abb7534cd4bc3b62527685dbd8 : SUCCESS in 7m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7893f1813c174ba19b5edc635443263d : FAILURE in 47m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dade0e7fff5b4cee8cb2779204f9128f : FAILURE in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d825d0442e0b3c92926c814817a7f479ced3e63b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-16 16:12:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/400/.","accounts_in_message":[],"_revision_number":1},{"id":"305fe0c3a4f5bd6e34e4df8d45f042d36daf6561","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-16 16:25:48.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_Unity_FC/5acb764/DellEMC_Unity_FC/1872 : FAILURE in 1h 35m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_Unity_iSCSI/02b4860/DellEMC_Unity_iSCSI/1900 : FAILURE in 1h 59m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5c1a7ae75f1f008fab1152a648e67b6161875268","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-16 18:24:18.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_PowerStore_iSCSI/8c16177/DellEMC_PowerStore_iSCSI/1322 : FAILURE in 2h 52m 27s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_PowerStore_FC/31cc7ab/DellEMC_PowerStore_FC/1506 : SUCCESS in 1h 03m 24s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_PowerStore_NFS/1cba402/DellEMC_PowerStore_NFS/1015 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_PowerStore_NVMeTCP/f517fb6/DellEMC_PowerStore_NVMeTCP/871 : FAILURE in 2h 28m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"453ddb4038afa73327932deded0458b946950eeb","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-05-16 18:41:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13756b93e3ae3304333a9a306e91218c31623bf8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-16 19:13:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_PowerFlex_3.6/cea9314/DellEMC_PowerFlex_3.6/1371 : SUCCESS in 1h 43m 34s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_PowerFlex_4.5/9769025/DellEMC_PowerFlex_4.5/291 : SUCCESS in 2h 17m 58s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/48/919848/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/5f38861/DellEMC_PowerFlex_4.x_NVMeTCP/236 : FAILURE in 2h 55m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"e4086c593e4fd6282c08d3b9c9dffa2ab6f457cd","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-05-16 19:46:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc77d832f63194889c02e788dbdff87be9a4513c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 19:49:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f28c72e9d469a31e4fe48c4ab2b8dd1057d9f335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 21:10:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b810ab11f09c47a3a141b468fbfdde28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe2ad0ae21749bf849f49731d2c926e : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/384ca28a5d5f4d8d8c1b8a82aa65daed : SUCCESS in 11m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8e2ba5bcb4e44d6968dd0216908a009 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8115cf86f1c543aaa22da70b792e1f0e : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c0e7280e78715fa8e917a17d20e7a68e4b414085","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 21:10:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"46190acde5519a6b1d3496544ab4f97897a4bcda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 21:12:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/399246be5cb74c2ca975fada41058e6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a986258f23844a9da6ce6dd25679228e : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f217d112fcc9ed30fcce56d768217f97e8231e48","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-16 22:58:38.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-919848-1/FC/ : FAILURE in 8h 32m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-919848-1/iSCSI/ : SUCCESS in 2h 11m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26a9884c2a29bd73bba3df733e9b964e521dd6be","revisions":{"26a9884c2a29bd73bba3df733e9b964e521dd6be":{"kind":"REWORK","_number":1,"created":"2024-05-16 14:25:57.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/48/919848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/919848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/919848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/919848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/919848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/919848/1"}}},"commit":{"parents":[{"commit":"e9c18c7f94357de32e08f9745de20fe0684a013b","subject":"Merge \"Support hacking 6.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9c18c7f94357de32e08f9745de20fe0684a013b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-05-15 19:37:30.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-05-16 14:25:41.000000000","tz":-240},"subject":"Fix fast8 tox env","message":"Fix fast8 tox env\n\nflake8 --diff no longer exists.\n\nChange-Id: Idf93686a13cfc40923bc142cad058c97fecf5177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26a9884c2a29bd73bba3df733e9b964e521dd6be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26a9884c2a29bd73bba3df733e9b964e521dd6be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
