)]}'
{"id":"openstack%2Fcinder~919868","triplet_id":"openstack%2Fcinder~stable%2F2023.1~Idaea18411ae03c5a9696b2ab67e84087177fcc74","project":"openstack/cinder","branch":"stable/2023.1","topic":"Idaea18411ae03c5a9696b2ab67e84087177fcc74-stable/2023.1","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-05-23 23:48:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idaea18411ae03c5a9696b2ab67e84087177fcc74","subject":"HPE 3PAR: use vlan iscsi ips","status":"MERGED","created":"2024-05-20 08:48:16.000000000","updated":"2024-05-24 06:19:04.000000000","submitted":"2024-05-23 23:48:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919868-Idaea18411ae03c5a9696b2ab67e84087177fcc74-stable/2023.1","cherry_pick_of_change":878684,"cherry_pick_of_patch_set":6,"meta_rev_id":"a3aac98d505145774b1e69b0554414db69953da1","_number":919868,"virtual_id_number":919868,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-23 23:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-05-20 12:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-22 14:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-22 14:08:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31804,"name":"Ryo Hayakawa","email":"ryo@redhat.com","username":"ryohayakawa"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-20 09:11:06.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-20 09:13:11.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-20 09:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-20 09:24:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-20 10:36:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-20 12:30:29.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":31804,"name":"Ryo Hayakawa","email":"ryo@redhat.com","username":"ryohayakawa"},"state":"CC"},{"updated":"2024-05-20 12:30:29.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-05-20 15:37:50.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-05-22 14:08:52.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"53b52016af6d3b8fac4ee01f4d254b226a29ef32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-20 08:48:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0883365a91f44277be2de83e791a299b1eb37448","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 09:11:06.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/24d15996a60b4d0197ca2c49db4d5fde\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/04fdacf0196b4887ae1dc5aff98dcc13 : POST_FAILURE in 19m 27s","accounts_in_message":[],"_revision_number":1},{"id":"57c4710e060f72192e8a31ffec598dca1efd6f88","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-20 09:13:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/919868 : FAILURE in 23m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7e66771a3f0e00e4dc9ef4a1fc9a6a7594b7521a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 09:13:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aa4930e323c44888c4b77d7df6cf9ee\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0292445b18d447b84a8ce1e628e2788 : FAILURE in 20m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8550e8f63a7ea6c2e397f79d495b41729c195c8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-20 09:24:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/919868 : FAILURE in 35m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ce6cb2b334c235ad6f5d4e3cff0513169317ef06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:36:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905a45dcf61f4649ace91e89a414bc12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4efc2342b64a90b0da0ee748ac8faf : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0ac27bd44dd4c46aadc17941c1dc9fd : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794b3729a96146ba89a970c233362e59 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/9d54346f4ba1455cb53c5983a2997e67 : SUCCESS in 1h 01m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/972981b012ee456882b09fb67a7b53f6 : SUCCESS in 1h 00m 03s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2741f7dd19034218bcd70f070ca1d0d8 : SUCCESS in 1h 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abbafccc60e549578c53d6467fb006f4 : SUCCESS in 40m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef3ff727839d40d79588cc32c60e3105 : SUCCESS in 12m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/93e59ef231a84a87ba5e909f30458fa3 : SUCCESS in 16m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8821c94d6c4f47a0b8e155d00feab044 : SUCCESS in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/254a8110d27a4146bff35511056235c1 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b095d32c17dd4f15afb0e4b1c1ff7511 : SUCCESS in 6m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88afd971d5d5454ab9170f275c32e41e : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ec027b041e4a4ab09cb73d8fc457ff16 : SUCCESS in 41m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fcd1527b56614ea196597c4a24e95a5d : FAILURE in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a87516477554213a96e1e25e3085793 : SUCCESS in 1h 40m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/340da837a18f478f8184b9b0fa2d21ef : SUCCESS in 1h 43m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6dbbf1bb9a6f4e349489e9802d518c3f : SUCCESS in 1h 46m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f6fecd3caa1348b7b423d526c7b3e4b8 : SUCCESS in 1h 08m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6144ed24418545e68103803df4bd1fd1 : SUCCESS in 1h 23m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/74c089a438404c77a0d4c9b76f4c94c0 : SUCCESS in 37m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a7d2b551852445a1949f35db1eae7f0c : SUCCESS in 1h 27m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a372d0437894d02bb34a31d997f005c : SUCCESS in 1h 12m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f42c53eb9f424fa2bfa0cb481666185a : SUCCESS in 44m 13s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4261564254d147f1979f08672400203a : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fd41f697b3eef44459dc83295ae2c9eabd63a74d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-20 12:30:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"555c172e4d4819a43dae1ed02b9953fb6559299b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-20 12:42:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d258a76d287a5ce2d2efe59c43d50e98969e8467","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-20 15:37:50.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-919868-1/FC/ : FAILURE in 6h 49m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-919868-1/iSCSI/ : SUCCESS in 3h 32m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c6741b339ff2e726db70c26d73ebb61b4064fd39","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-05-22 14:08:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"179b7ef260c7256c007faf6d7f907de69129f25a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 14:09:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f29f743248a0b51d3cda13bd07178c81f6efd374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 17:08:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24f3378eee234ce2bb595823e73614de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90df9d6ced354ba38554e92fcc5366d8 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/929e145dd5bc499a96df78eed98a7d01 : FAILURE in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaca4290937f4a2cb530feb2c4f66850 : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/a43c0566af1e451cbfe9e07939bb61aa : SUCCESS in 59m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/336441683b734e41b5b389b3b8b0cd1f : SUCCESS in 1h 21m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecdb488ed46f48f2b2de1ce78f14213b : SUCCESS in 57m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23f9a044b479439ea8f33a20729394d1 : SUCCESS in 15m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/184988dc061f4e299bbf4b68cb3170f0 : SUCCESS in 1h 14m 24s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0f21d57a39634e0d860f777a0f6b10f3 : SUCCESS in 1h 29m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c3d5086107243bcba7a080609d98176 : SUCCESS in 47m 54s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b82ef1eb796414eb05e86b409c1d6f8 : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":1},{"id":"701de0017cf0b1aadb675331243c2acff121fa4f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-23 13:23:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"069e66e7fe9b90139364f2de951d28d4d45b44f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 13:33:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40fedb90238a442c90a5c1d318659123\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8aa5b6b702124086b5f241f155e3d0a5 : SUCCESS in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79136acab2182794be9195fa87d98d99f8413908","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 13:47:01.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b82874d9109944ce99b0d4be0e0bf66e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f1497895448940c19551bfbc217e5f96 : POST_FAILURE in 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"654392839c0302e0e3c4a8346e175c3b0dbbede1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:49:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4e5f04c1e8543cdb26b762a3e2b46c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eb652683f014324a8c27f1f1403d1f9 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92942588ff9d41e0a46dc7acf52eb116 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9c62b18e4b49f28090b34ff004a187 : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ded848ee5b4248b59920caa83f4ebf12 : SUCCESS in 1h 01m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a4b97dae0f1042e2bb75bc7bb10088e2 : SUCCESS in 56m 53s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/828ea249777f47328093198fb601305b : SUCCESS in 1h 15m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e930b3e7d794b9c9a3ff5e919cca9f9 : SUCCESS in 58m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20bc06b999234a5a8b563a3d35d797c6 : SUCCESS in 11m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0747c520a2bd48208f06ac3fd4a9c12f : SUCCESS in 12m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc1198234c824adb990f667829698f09 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/71cf00a6e73245eaa934bba5afd64f79 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cbfb1e969adc498c810bc14bacc1fa52 : SUCCESS in 5m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cad561a8acf343ba90857328bf2450d0 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7132d298855413b85949880474aa189 : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c940ef35e58e43f18669ba9cdfc8bdb2 : FAILURE in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/50d1b4ac9e9c4e609921a3e542bb1d76 : SUCCESS in 1h 56m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04df63f5638c4862b96e7bb09ce9a943 : SUCCESS in 2h 18m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/caf53cac732c42598db7f87b5c0c92a2 : SUCCESS in 2h 02m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4961797707744b6eb7c51ce2fd9c00fa : SUCCESS in 58m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/28349963ba0047f1b4d6ab0d7de81b95 : SUCCESS in 1h 36m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aac3c6c6673848d6b685002ec52d9102 : SUCCESS in 43m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e5ab1efb04284cfd8392b9cccd402603 : SUCCESS in 1h 16m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df388ef28e8047e6909b861838ce7da2 : FAILURE in 1h 16m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15ac0a85f7c343b5883ae7f2f27698f7 : SUCCESS in 51m 40s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37d795e8760240d09a319fa27038cdd7 : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c459d7b5ff2464dbca29e4e849f655c594ea42e1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-23 20:21:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0b0a4def66b011425f386cd2c2ab68b6203ed5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 20:32:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77f92229f17499b9c68ffe4466da863\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2147f1c6c1384ecc8f817d68e231d6aa : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08e49526508b047c1d67222f57616090cd659134","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 20:47:40.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/621eb03510f04c4cb927352ca22de6d6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d2d4c70d8c6c4ccc9d4d00640f3ded11 : POST_FAILURE in 21m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cc36fa859e39e513230c706ef8d5c9f6124abb67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 22:24:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/138f13ffb75145b6a65656487e9b5b93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0154c79a3fdb41d0b7850fc8359f307b : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5443e71fbbd14b409df842edac5ed110 : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9f9ecd4742478a9eef48ce5dcab7eb : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/7b0d71b30c9c4e4986609782b509af4c : SUCCESS in 49m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a63093f39580407e9bfae44cd6a74acd : SUCCESS in 46m 23s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/82b5d4259f99408a8c48e18b2526c9b9 : SUCCESS in 57m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/386f80a64c5b4afbb7ce279524635bb2 : SUCCESS in 41m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33d3969597e44015aafd81424d61b8fc : SUCCESS in 12m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/be3db8af109c4f778e602e37b03b16ef : SUCCESS in 14m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca35688a7b5d461abea60ab52d7c4bce : SUCCESS in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/13416b091fc642db8fe45192bc499269 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65d4b79c418947fcb2c3b866816e052c : SUCCESS in 5m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/54cc181d63d3427ea205f8bc9638cd4b : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e8a9844f10c5457e92d538c6d14957d3 : SUCCESS in 36m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8afd506c8da64442a627baaed3ff8070 : FAILURE in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c2d6606f83174bd4b31280b45fdf64c5 : SUCCESS in 1h 47m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/60c4cbf6325342f79205778ac2775b1f : FAILURE in 1h 56m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/55346e99811240329399ce8ada280940 : SUCCESS in 1h 23m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a78c2cd0e2714de9a02e7a2d11ded458 : SUCCESS in 1h 07m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0090be653b054122a069ada1b09e612b : SUCCESS in 1h 36m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/90f6bde4fa8b41e6ada953f922248ce0 : SUCCESS in 37m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3a1717d73c6a41ac8e731f7489bd5495 : SUCCESS in 1h 07m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/31e20810d5404cdbaed4573166f28992 : SUCCESS in 1h 04m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59720d126f8443c6ad59f47e2ac05968 : SUCCESS in 52m 40s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d33a5893fa984f9a9e5aa42a1161e9d6 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ebcd4b5d0a357965ec5c2d7fdda123f781f2ae9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 22:25:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49014d5df73f647686623df703504b895e71d21d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 23:48:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4732e01e2eeb4f3db494f8289d4d1def\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13d37018870746e5ac5d2add92256fb9 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d450ea8f987c496e8c2ad1ecf6136795 : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be57838b6cd0493db16b557dec5d1e17 : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/212bbec4ac7f41a7b77ec44d4ac6f21a : SUCCESS in 1h 00m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b56e3f79ff87415fb5341263976586fd : SUCCESS in 1h 23m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dfe65513e504d8d9257c62fa5cb6e13 : SUCCESS in 55m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f91a66e96ae1465b8494118b9fc34fc3 : SUCCESS in 15m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/454080dfb40d4ce39829e6d5c2e04bc9 : SUCCESS in 52m 35s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91d6a1dcd4684f9a80b242509eda9c11 : SUCCESS in 1h 18m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/247a1fbdae0c47099a56cbf6235c9f9f : SUCCESS in 50m 03s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f10f21a8c584d78903eeda23c8742e6 : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8c69bfcb661e8ce8dcd4e9d215e82f5cb9cba737","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 23:48:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a492b5d2afc846d5ca847755501e901c3bdcce92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 23:50:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e40dcc5286834ab5b5f19c1349367783\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2bcdb2cd9d4118b52aef49757b856d : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6ed689c728e448eb0a67e6903cacaf5 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"a3aac98d505145774b1e69b0554414db69953da1","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-24 06:19:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b30ed079fc4073d38f9b52a2f9ae497d4967413a","revisions":{"b30ed079fc4073d38f9b52a2f9ae497d4967413a":{"kind":"REWORK","_number":1,"created":"2024-05-20 08:48:16.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/68/919868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/919868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/919868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/919868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/919868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/919868/1"}}},"commit":{"parents":[{"commit":"0cca8c2aaf20cea530278bc481e94d647de1793f","subject":"[stable-only] Set volume_image_dep_tests flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0cca8c2aaf20cea530278bc481e94d647de1793f"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-04-03 05:26:41.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-20 08:48:16.000000000","tz":0},"subject":"HPE 3PAR: use vlan iscsi ips","message":"HPE 3PAR: use vlan iscsi ips\n\nIn the code hpe_3par_iscsi.py, only the trunk ips\n(192.168.68.203) are considered. If vlan ips (192.168.100.1)\nare specified, they aren\u0027t considered while forming list of\niscsi ips.\n\nSample output of \"showport\" command from HPE 3par storage system:\n\nCSSOS-SSA04 cli% showport -iscsivlans\nN:S:P VLAN IPAddr        Netmask\n0:2:1  -  192.168.68.203 255.255.192.0\n0:2:1 100 192.168.100.1  255.255.192.0\n0:2:1 200 192.168.100.2  255.255.255.0\n\nKindly refer launchpad bug for details.\n\nThis patch adds support for using vlan iscsi ips.\n\nCloses-Bug: #2015034\nChange-Id: Idaea18411ae03c5a9696b2ab67e84087177fcc74\n(cherry picked from commit 4482ec892f00a724fb2b2a56607d0815fe0c17e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b30ed079fc4073d38f9b52a2f9ae497d4967413a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b30ed079fc4073d38f9b52a2f9ae497d4967413a"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
