)]}'
{"id":"openstack%2Fcinder~920006","triplet_id":"openstack%2Fcinder~master~Ib50a78a549d3bd20f98e4b51298b5ce9bd276b10","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"Ib50a78a549d3bd20f98e4b51298b5ce9bd276b10","subject":"Fix: [Ceph] Restore to a larger disk than the backup","status":"ABANDONED","created":"2024-05-20 03:03:08.000000000","updated":"2024-05-20 10:50:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cf0e21b63914bce1858393613116c22c02d12078","_number":920006,"virtual_id_number":920006,"owner":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-20 04:23:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-20 10:50:35.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"}],"messages":[{"id":"5f61c7f97a0204b39d0930140b335061d3a76cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"date":"2024-05-20 03:03:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"882d981a514660700f04cfa200ba73c6b336b516","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"date":"2024-05-20 03:12:49.000000000","message":"Abandoned\n\nThis code review is a duplicate of https://review.opendev.org/c/openstack/cinder/+/791059","accounts_in_message":[],"_revision_number":1},{"id":"df950d023fd8d08d6f60fac49f893713ad4c0498","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-20 04:23:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/418/.","accounts_in_message":[],"_revision_number":1},{"id":"cf0e21b63914bce1858393613116c22c02d12078","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-05-20 10:50:35.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-06-920006-1/FC/ : FAILURE in 7h 46m 59s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-06-920006-1/iSCSI/ : SUCCESS in 4h 29m 51s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"560c11f0ac4b940987bdd4e26f492e2ecfaf24fa","revisions":{"560c11f0ac4b940987bdd4e26f492e2ecfaf24fa":{"kind":"REWORK","_number":1,"created":"2024-05-20 03:03:08.000000000","uploader":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"ref":"refs/changes/06/920006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/920006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/920006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/920006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/920006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/920006/1"}}},"commit":{"parents":[{"commit":"0fbd736183191a9d4516ebb8223beb8af317c4a6","subject":"Merge \"Remove the duplicate code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fbd736183191a9d4516ebb8223beb8af317c4a6"}]}],"author":{"name":"Ber1in.Wu","email":"wbl1652124020@gmail.com","date":"2024-05-20 02:54:10.000000000","tz":480},"committer":{"name":"bolin.wu","email":"1652124020@qq.com","date":"2024-05-20 03:02:52.000000000","tz":480},"subject":"Fix: [Ceph] Restore to a larger disk than the backup","message":"Fix: [Ceph] Restore to a larger disk than the backup\n\nWhen restoring to a disk older than the backup, you need to write\n\u0027\\0\u0027, but the current syntax needs to be compatible with python3.\n\nChange-Id: Ib50a78a549d3bd20f98e4b51298b5ce9bd276b10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/560c11f0ac4b940987bdd4e26f492e2ecfaf24fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/560c11f0ac4b940987bdd4e26f492e2ecfaf24fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
