)]}'
{"id":"openstack%2Fcinder~920073","triplet_id":"openstack%2Fcinder~master~I367df9b52f787403a54fb148713d122e9bb2a54f","project":"openstack/cinder","branch":"master","topic":"bug/1925809","attention_set":{},"removed_from_attention_set":{"36978":{"account":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"last_update":"2024-07-25 14:11:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I367df9b52f787403a54fb148713d122e9bb2a54f","subject":"Fix: [Ceph] Backup Driver Python3 Encoding Issue","status":"MERGED","created":"2024-05-21 05:50:35.000000000","updated":"2024-07-25 15:51:59.000000000","submitted":"2024-07-25 14:11:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"920073-bug/1925809","meta_rev_id":"5a3bf7b5f393b2b01a3e83c47978fa971a06fb0e","_number":920073,"virtual_id_number":920073,"owner":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-25 14:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-07-19 12:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-07-19 14:46:48.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2024-07-19 15:09:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-07-19 14:46:48.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-21 06:03:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-21 06:05:44.000000000","updated_by":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-21 07:45:04.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-21 08:25:15.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-21 10:01:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-22 08:19:41.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-23 05:46:04.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-23 07:53:01.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-23 14:41:39.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-23 23:47:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-28 04:33:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-29 06:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-05-29 14:58:17.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-07-19 12:58:58.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-19 14:26:13.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2024-07-19 14:46:48.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-07-19 15:09:13.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-07-25 11:20:30.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-25 11:31:21.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"f07a6faab3e498f6fccda79ea827d2de75958755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"date":"2024-05-21 05:50:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d6dce2a2cfcae39d1cf9f17e77cba9b634d7c3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 06:03:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cacf2b7da0d54c84afe19c3733a96d14\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04e7c24835344276b2c257cb2251ed06 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/770005b918e54e688231068f2a894d9b : SUCCESS in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"373d08930333d77ed51df86b78e65ba9d380dcd7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-21 07:45:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/920073 : FAILURE in 23m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e570abfb73f48a521787e199d06b4a62acc1bd95","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 08:25:15.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a622d3cc662c4611aabac615ea9df24b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ad0631fbd3024ae4b77c50bb8b72e78e : POST_FAILURE in 22m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8c065ae66654474cebab5c433919ab2c396beda4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:28:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d14f12280f684d318c040e88591cf4ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ac680636ba45c6bb93f9db1bc0ed7e : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4d5ea4b7f394e16b9430991df1afdbe : SUCCESS in 12m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f41a58635cad4172914b0b2347fa2c3b : SUCCESS in 10m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9c171bb49a34d80bef0cb067a64509c : FAILURE in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebd209bbd7404a80a51cc353fdd9bc99 : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/06560a2244104cbf89763643f0d8b650 : SUCCESS in 1h 03m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f4ffc23a2964d19aff5a2eac4f731a2 : FAILURE in 1h 17m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/650b77a69eb74153ba46da289ad27bc2 : SUCCESS in 56m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a52f68a4815d47a99983b70c938bf5e4 : SUCCESS in 16m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f226b97ef1d3482aa1d519b085c0fe02 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb7f8d63525640868ac900e1171488c2 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a2523f54d1349e4b9ec99378558b290 : SUCCESS in 5m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fb200e54b2594cce9cf6a8cf0f166738 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4fc80d105b4343278f166d81f5fa82b5 : FAILURE in 44m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5bb79aa8859c4c8887e3883e069d5002 : FAILURE in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4bea5b28a99849f7afe6be76012f00a3 : SUCCESS in 1h 53m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bd93c49dbbdc4f12bce0759efc71c715 : SUCCESS in 2h 05m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e2df3a09b0e44389867844e6da0d9b78 : FAILURE in 1h 55m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0e163ef4a8a1475c8aac0f666fac97f9 : SUCCESS in 1h 54m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d6a1f12536ea4d14992138a701cb22e8 : SUCCESS in 23m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d810ba3cce3042a7a031ac37f0adfe43 : SUCCESS in 1h 13m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/77fdc27a7c014383be2ab122919a23bc : SUCCESS in 1h 40m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/00cd86622c714c5993916d9aacbbb131 : SUCCESS in 48m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9df1db26d08c44489140bcc268830804 : SUCCESS in 1h 26m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4a3b84670ae54186856a811467241a20 : POST_FAILURE in 1h 44m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e8429fddc91644569cc0683993b84a82 : FAILURE in 1h 26m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6072956a0cba49569eca7af4bfe3b2a2 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ad904bed6e678802b6976040760bea3e3dcaf053","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-21 10:01:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/920073 : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9b036078b1f7f18b3c8258b3df2a5598e9b66487","author":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"date":"2024-05-22 03:30:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"deeb535b98ec6e11274bc56455110959183288b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 03:44:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d50073605a445fa43c2827b8e9aa83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c98f7b569c34c5e97aa54aed83bca2d : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/456710ae30944f418798df97fe28cfc6 : SUCCESS in 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccb9f90bf09ff20663b27d98af8269a77c9b15b3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-22 03:57:32.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3b1765ba36bb406abebaeea0132ed520\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f7b36b18f9374c3ea665fbb07bb4e2fe : POST_FAILURE in 21m 01s","accounts_in_message":[],"_revision_number":1},{"id":"477dae767039800c09ab5e74a63425c07149de33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 05:52:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b1399e574c446c4b8aeeb5db93d0bd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffc0b1c3dd043e58c24ff9ad9e03e37 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b0d8a6e66034602a5ddf99a2053d2fc : SUCCESS in 11m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e0faca1e7394170b13ee3bea8dd22a0 : SUCCESS in 11m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a032e235a0b491bacdefa40b56a1bde : FAILURE in 13m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b11ad32fce4aae985ea2cd95df832d : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/39cb812194ed4fa8886f89994f351c96 : SUCCESS in 1h 04m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ee0124c2aad450594b4138ab6e36c52 : FAILURE in 1h 20m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5517fb4deee545d1927015833b8e126a : SUCCESS in 1h 06m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1769638e892b4565b8fdf94e118a1251 : SUCCESS in 20m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4f0436d3117a4b3fb84f1abc49631aac : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8976aca58514b3989efaa95523a238f : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c781ec1167341eeb51616410efbcc7f : SUCCESS in 6m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6bbf9dd6b6844e86963248f741e4547b : SUCCESS in 5m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f24a8ba3fb1348ffa7dd448a081b8e5d : FAILURE in 35m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47fa1b2273fc4ce58b99bae57f68b83b : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58f8576fd8cb441091cbc66843e03482 : SUCCESS in 1h 55m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0805b04b692f445782909f193831a713 : SUCCESS in 2h 02m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1acb174368e54dcab8792db99c673a4b : SUCCESS in 1h 39m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ca32e3a8146f4746a979ab2a041ea695 : SUCCESS in 1h 30m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/26a38e05ae264d809e46338589b27adb : SUCCESS in 33m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/afa18d0d8c214835893a28339afda4b9 : SUCCESS in 58m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fdcb07eb26a44efcb6ddf1211a52404c : SUCCESS in 1h 26m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/426e92be538f4851917bb546c7cd94cc : SUCCESS in 34m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d6a6a1fdd88d4152a0b01b57d1d75128 : SUCCESS in 1h 27m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9cdd5eb1457b45e9aa5d2309bee483ef : SUCCESS in 1h 31m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/273471e5eebd4827a40ddfd2383ac127 : SUCCESS in 1h 20m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef7ca70ce1df4df3bd59c158b31d69ba : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2eb685ffaaed3595a480cafc4b570a322b45f3cc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-22 08:19:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/454/.","accounts_in_message":[],"_revision_number":1},{"id":"a3b444ef1ad250492f2d8412fd7b8c5301efebbb","author":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"date":"2024-05-23 02:51:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b395f3a8907768cf5048cff557da7662c733b498","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 03:14:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b34f2e417fa34f81841b1384598d1a6d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0dc8414d4a38404ead7cfc499f45b1cb : FAILURE in 20m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50cf14a4e3e04c0cbab4c5055db99ac0 : SUCCESS in 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59945907f476f9def5fb728ceeedbd197ef81aa8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 03:20:14.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/bf421876366e47009c1b1152b6bc61f1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4ffd2efcbed1402aba306071c7613232 : POST_FAILURE in 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d048724b6f4b88868f9718e8efed08712ad1dfb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 05:10:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95bb53cb5c8f47ad89934bcd8163b674\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246e163070d242d3a4de522a85bd7267 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ecb47088f604308bdcdae933cb47c66 : SUCCESS in 10m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21e0a829ed7d44d5b3d1f4d486f10831 : SUCCESS in 11m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fa25d1fab614b3296fb2b6f473c9b16 : SUCCESS in 16m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d9a06abd37463fb0c1667581284139 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/ea7b4f23e3d7496ab77135282243c4ff : SUCCESS in 1h 00m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8d9f1e8181d46eca3f2b66abd9a0751 : SUCCESS in 1h 33m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9edcfdd3b2cb4fe3bf5dd135b2426f13 : SUCCESS in 1h 03m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f352db9b47d40f99cac778106581c5f : SUCCESS in 17m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8a148b8e0d7c461f8c3cc47387694aa5 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d9bb0bceb7749afa59c5fde7e113eca : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51aee3324cea4aefb1719911a6744d46 : SUCCESS in 5m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/916e38c446cc491380edb9fd4a7125e0 : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a42afca7df947fe9dbb701c276ea240 : FAILURE in 32m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef1a0f9eccac4cdba51e2e29edad3e46 : FAILURE in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/990bd910f1cb46cfaf11ce826b64b4eb : SUCCESS in 1h 50m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/51560c0da0b84f8c9873709bcf87f9ad : FAILURE in 1h 52m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cd35c69ec8bb4fdaafa0c8b3b8affb5d : SUCCESS in 1h 18m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b7ca9017dda7474cad23721e2594cdfe : SUCCESS in 2h 13m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/23d69b96c60a4e29909a09bcda8fe201 : SUCCESS in 35m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79d011855fc94ba1a7f2a0e590bf9008 : SUCCESS in 1h 02m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ab83a5746dc64e39beb78baa2cfd79f8 : SUCCESS in 1h 29m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aed3aaaefde94daf9eb45630853b3964 : SUCCESS in 36m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1c44d073126441ecb36d1a485d8df592 : SUCCESS in 55m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b932f430aad049578035565daf1b0147 : SUCCESS in 1h 14m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f1e8437a6b7a4d58bfea76c875eb7f85 : SUCCESS in 1h 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad50cd418f144cad944d91e1fb14b333 : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ccdf939b3e27ceb3fcebfa0ba19819d8b3744e11","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-23 05:46:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/920073/1/upstream-check/cinder-tempest-plugin-ontap-fcp/1cda8ac/ : RETRY_LIMIT in 3m 32s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/920073/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/5a92612/ : RETRY_LIMIT in 3m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/920073/1/upstream-check/cinder-tempest-plugin-ontap-nfs/f7e241a/ : RETRY_LIMIT in 3m 39s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/920073/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/eb0062a/ : RETRY_LIMIT in 3m 07s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e78c1c6a4593f9d290dc5f7327407c72f85c34fe","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-23 07:53:01.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_iSCSI/c877f1a/DellEMC_PowerStore_iSCSI/1376 : SUCCESS in 1h 22m 07s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_FC/17a4391/DellEMC_PowerStore_FC/1559 : SUCCESS in 1h 01m 11s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_NFS/a46663c/DellEMC_PowerStore_NFS/1070 : SUCCESS in 1h 08m 52s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_NVMeTCP/ab26356/DellEMC_PowerStore_NVMeTCP/924 : SUCCESS in 1h 35m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"21af7e7e99e1da19ea4c2051114160169acc5766","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-23 14:41:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_Unity_FC/1357ab7/DellEMC_Unity_FC/1934 : FAILURE in 2h 11m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_Unity_iSCSI/6bdab15/DellEMC_Unity_iSCSI/1962 : FAILURE in 2h 10m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"516a438d0e39e23d1a30603b6777845780b4ef2f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-23 23:47:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerFlex_3.6/be2a658/DellEMC_PowerFlex_3.6/1433 : SUCCESS in 1h 42m 28s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerFlex_4.5/e19eae0/DellEMC_PowerFlex_4.5/347 : FAILURE in 2h 06m 38s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/622852d/DellEMC_PowerFlex_4.x_NVMeTCP/287 : FAILURE in 2h 55m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9aaf6aca0182062636c88c432a9adc2a1363b178","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-28 04:33:52.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/920073/1/20240528053328/NEC-FC : SUCCESS in 5h 26m 32s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/920073/1/20240528053328/NEC-ISCSI : FAILURE in 8h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0e3063e1088cb3545e9b42d06a0e687705e95038","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-29 06:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/73/920073/1/fc : SUCCESS in 0h 45m 08s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/73/920073/1/iscsi : SUCCESS in 0h 52m 02s","accounts_in_message":[],"_revision_number":1},{"id":"96ba969630762bb31d3793447501c855dfe6bfad","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-05-29 14:58:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/920073/1/20240529200509/NEC-V-ISCSI : SUCCESS in 3h 22m 28s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/920073/1/20240529200509/NEC-V-FC : SUCCESS in 3h 53m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b3c063b9c0c1c6a1e8dcf69da20687850262be28","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-19 12:58:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62ffbd58993d3e574e6ccb1739f1e14bb9c3b775","author":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"date":"2024-07-19 13:36:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9aeabea1fa10f2ce58f8933c7a418e662ff0cbd","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-19 14:26:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eaeae01ed4f2427026ce0e2b901711b86fe34d09","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-19 14:46:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad3522f3517cada152facc4f7695141992eb8b7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:47:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d45efc31d65fedc54906bb3a020b726af2a47d7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-19 15:09:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d9222313c745fc4d4c36f28a8085cd7e2640ac08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:50:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c026ace71617480ba3b63039b7a93371\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96edb8100f6b4f179696b767b3b0fdf1 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9117cee1cc64c8b8f12071ea7be313c : SUCCESS in 10m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0125e2552f3140f38fe4aae93afb3873 : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55489fc7785048a282779f1649d02632 : SUCCESS in 16m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/696d74ae61d24c3c8a4ffa63512a1599 : SUCCESS in 58m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/535edc343e964fd8a6eec18e66ef4ce7 : FAILURE in 1h 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d3d3486dd984b279deb2f46f28e2e5c : SUCCESS in 1h 02m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e64465c72b3b486785b978d1009b0f09 : SUCCESS in 1h 11m 06s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a804a60a659a4a64866b1166d3212ae7 : SUCCESS in 1h 51m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef6c1e06e02e443d8efe01bb13a0d3c6 : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bf389a80690de5bb277af18e7dfaa78267c42f42","author":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"date":"2024-07-25 09:03:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c796292ea933800382be70809107a1e594d3c5f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 09:24:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8bf8a0a08e947708d24bad2ae3f06a9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59988abb7bd74dd7b57882e75e500caf : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/551bc287b3bc484ca11176c6aef5514d : SUCCESS in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0854b87fb1438cb149d8e7edd874a4efe2231ccc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-25 09:32:43.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d40d35f3d7564ebea457dddf7cb1c584\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/87d4d232559f46ccacf16033a4b1c9ed : POST_FAILURE in 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"eddceaae01bba31c352944cf012c65d24d704e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 11:14:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac31aca403842cd95f82adda03a592f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679680c62bce4994be1db26d08f3674c : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e74923c8fda4819b67789fc7d09bc11 : SUCCESS in 9m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff4ee004439c4e4e82541e11068f2737 : SUCCESS in 10m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cb6d89ebd0c45559d3fcc23704e399e : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3382de7f42f34beebfbe2a036409f5cf : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/4f0db1eca3ff4324a045e206be498f02 : SUCCESS in 48m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/537f46a71ab9414da72d7aa4a27808eb : SUCCESS in 1h 21m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15c99f44d6714773a22d62e2925f0025 : SUCCESS in 1h 03m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84ff7395af214043b9c3c445c27f16f5 : SUCCESS in 14m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/551c3c77097a4599973be2483983eb33 : SUCCESS in 5m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c8b12b61390a439e8c037536f7dc6909 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8fb019fe117549de822039276307e85d : SUCCESS in 8m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1869b6518fa0427e98fbfc4f15e0c8e4 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f5904805fc734be98ea3f02ec3eb93bc : FAILURE in 41m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a519c34677e4c289fd4853cbb10a116 : SUCCESS in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b818e5dcccfa4301b85650353615d96f : SUCCESS in 1h 32m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4495b97933044d5a92183a9b4de23827 : SUCCESS in 2h 07m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5770a8442d3e461f921a400ae1553c40 : SUCCESS in 1h 54m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/294b593e85824a14a630208f0dca7c3e : SUCCESS in 1h 50m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/66635285977b4aa5a247dc61ba9f88a6 : SUCCESS in 33m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/270677d4897a4f70ab96ecad5fe1b067 : SUCCESS in 1h 06m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f3ea13739c504c48981363df58b69a79 : SUCCESS in 1h 12m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/27faae736b0e49f2849952672bde8195 : SUCCESS in 29m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7099ec3a7d0944b2a4accb66dd0b19fa : SUCCESS in 1h 24m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d9ab75f748844845bc4e2a1b85eb308a : SUCCESS in 1h 39m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d1c9f3cfb404babb104461ab7f8ceb6 : SUCCESS in 1h 40m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/facaa835faec49f0bf491080274ae430 : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9bba4a9ac6696fb74a2ce6960798f028fbf4ddf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 11:15:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aa1ac15d1a1fd93523cbb37ce9c8584956b46ae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-25 11:20:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_07_2024_11_14_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"12992730ad1f010f0bf4d875595b546480b7bc9d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-25 11:24:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/920073/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/5044b69/ : RETRY_LIMIT in 3m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/920073/1/upstream-check/cinder-tempest-plugin-ontap-nfs/13fcbd4/ : RETRY_LIMIT in 9m 06s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"0b376565852fc963e9b8497a19e4b3e4584bde20","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-25 11:31:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2fe1ad4d8fd544f6a1463fb632bf38ed : SUCCESS in 1h 13m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/09f07f95da1248a5b15dbac66574d310 : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ced2940c6996688577c28ca549f1607763a83c3d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-25 13:14:57.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_iSCSI/950e279/DellEMC_PowerStore_iSCSI/1541 : SUCCESS in 1h 21m 58s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_FC/5cc1e0c/DellEMC_PowerStore_FC/1737 : SUCCESS in 1h 04m 04s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_NFS/6ac4bb4/DellEMC_PowerStore_NFS/1235 : FAILURE in 1h 17m 08s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_NVMeTCP/19eef82/DellEMC_PowerStore_NVMeTCP/1094 : SUCCESS in 2h 00m 07s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerStore_NVMeFC/3112781/DellEMC_PowerStore_NVMeFC/58 : SUCCESS in 1h 12m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"38a0a1e7233f865644ab44398f058e12bb134cac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-25 13:36:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_Unity_FC/020ec0b/DellEMC_Unity_FC/2296 : SUCCESS in 1h 53m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_Unity_iSCSI/1ba6593/DellEMC_Unity_iSCSI/2303 : SUCCESS in 1h 51m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6fc2d18cd13f66debb78633f577d27542908ea96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 14:11:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab8f85ed33d469c8edec20e96e93ce1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87081bbc72249ed9dbf7c961a60168b : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cee5d4cf29c84b7dba9d4a5d9afe496f : SUCCESS in 10m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97e7956592d9443b8770b9fd399615ef : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3c022a5ac54905b23ac6d372f1724c : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/0c00f2b794724867a276d7778a65a889 : SUCCESS in 1h 08m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/25b1baa4b86c429eb1ab18b5517d1f98 : SUCCESS in 1h 24m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4d5a425672d472e83bbb285f2f81df4 : SUCCESS in 51m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/93b079070cfd430bba7335ddf19ab9d8 : SUCCESS in 59m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0392630cce644d1b990f54c59af8992f : SUCCESS in 1h 52m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ffaa9a0b4c34ccea39167ddb265573a : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f81c089ac47eb37afa7ce5f7feca2bb41ebe0893","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 14:11:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"53cc463d135a842043550cc61987e2fcaa3c2bce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 14:12:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec43cbba3be34745acd92f90db79956a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72897d3e492d4c0bb4416172b152557d : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5a3bf7b5f393b2b01a3e83c47978fa971a06fb0e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-25 15:51:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerFlex_3.6/72f1e36/DellEMC_PowerFlex_3.6/1784 : FAILURE in 2h 49m 44s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerFlex_4.5/ba70356/DellEMC_PowerFlex_4.5/663 : FAILURE in 2h 15m 35s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/73/920073/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/885fb47/DellEMC_PowerFlex_4.x_NVMeTCP/621 : FAILURE in 13m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c13370cdec32ae6b74f3653981519af567be40de","revisions":{"c13370cdec32ae6b74f3653981519af567be40de":{"kind":"REWORK","_number":1,"created":"2024-05-21 05:50:35.000000000","uploader":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"ref":"refs/changes/73/920073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/920073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/920073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/920073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/920073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/920073/1"}}},"commit":{"parents":[{"commit":"ccff71a251a68c0bcbc1a38f6e8435cddacf33c5","subject":"Merge \"swift backup driver: Ignore 404 during object deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ccff71a251a68c0bcbc1a38f6e8435cddacf33c5"}]}],"author":{"name":"bolin.wu","email":"1652124020@qq.com","date":"2024-05-21 05:47:33.000000000","tz":480},"committer":{"name":"bolin.wu","email":"1652124020@qq.com","date":"2024-05-21 05:49:47.000000000","tz":480},"subject":"Fix: [Ceph] Backup Driver Python3 Encoding Issue","message":"Fix: [Ceph] Backup Driver Python3 Encoding Issue\n\nUsing python3 the Ceph backup driver faces encoding issues.\nIn order to solve this we should start using bytearray for\ncreating a zero-filled instance with a given length.\n\nCloses-Bug: #1925809\nChange-Id: I367df9b52f787403a54fb148713d122e9bb2a54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c13370cdec32ae6b74f3653981519af567be40de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c13370cdec32ae6b74f3653981519af567be40de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
