)]}'
{"id":"openstack%2Fcinder~920326","triplet_id":"openstack%2Fcinder~master~I64d16f0da2587f392ef676857950c1de897414d2","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-06-03 06:18:20.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I64d16f0da2587f392ef676857950c1de897414d2","subject":"DNM - For test only","status":"ABANDONED","created":"2024-05-30 06:14:07.000000000","updated":"2024-06-05 10:51:50.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c6bd4cbf241b2af96a3078b08313105869c57dd","_number":920326,"virtual_id_number":920326,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-30 08:54:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-30 06:26:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-30 06:37:17.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-05-30 07:38:07.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-05-30 08:21:31.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-30 08:54:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-30 08:56:51.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-05-30 09:01:14.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-05-30 09:08:00.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-30 10:09:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-30 11:08:42.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-05-30 22:11:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-31 21:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-06-01 15:33:52.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-06-01 23:10:02.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-06-02 08:16:07.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"b9c3af04d93f092b59a52582378c6ac64887577e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-30 06:14:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89e0451bc7c951a35c0f5ce8643f6af796caa0ab","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-30 06:16:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d84094aa6e3184383b14803857e97ce106f4611e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 06:26:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69286f4b4c3471d9278c843b8a408b3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/927880be50c44c588d6c5cf49a18e2b7 : RETRY_LIMIT in 1m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ebb43dbc89e641b6a78ffbb53fa8bcfe : SUCCESS in 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f383fe3e430667b05423d0f2f14000dc505ce67","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-30 06:37:17.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/97bfdb9b97d2409594ed4e8740ab8eda\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1fe74b681b244063990ce9c2f6c5d08d : POST_FAILURE in 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7742c236f615fa3449dd99e255f27920f86d526d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-30 07:38:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/920326 : FAILURE in 1h 03m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7eec8b1bcf337bfa25121bd4eb45f0d070d0f8c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-05-30 08:21:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/354bdfe0ae57433ebac68621b52cd465 : SUCCESS in 1h 20m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eca89d96c80b46b3a1fb8d6e4d6311e3 : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":1},{"id":"433e82e7cecae77194b350c909b6eccdcae95c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 08:54:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/769047c514cb456181a0d0b48f4d42b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12ce04714d9420d985ce01202d80e05 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d53a72fbc14f490f8598ee981a154e7a : SUCCESS in 11m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a88aa74835ff41c08b7b010ec0323bd0 : SUCCESS in 13m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bfe2d48b30a14270a3a3d88622636172 : FAILURE in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75d10921a9c42af8d6f121797937514 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/bcb7246f0b58462cb784b12c20616353 : SUCCESS in 46m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/adfdb23ebf944f9294330682eb33ba6a : SUCCESS in 1h 50m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/185b7e0bb4b34e2ea08b73bb540b21b1 : SUCCESS in 53m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/191a93ac6fa14d0f8b3f7a52370508bb : SUCCESS in 17m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e84ff3120d0451393a207dae9553141 : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0da0cceef9ec4722a85ec6f443b489b4 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/02a703d618aa480f94206a545a30553c : SUCCESS in 6m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dee909bc5a0743dd8a8c94fc22a9d57b : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/160d1f16af7542dbb227f49815666a67 : FAILURE in 45m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/273b5424bea346dca000c68be7ce8dcf : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/810ff813b4c54e238e0677cfeb8009b1 : SUCCESS in 2h 17m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5d9fbe02d86d4958bd42469320c163de : FAILURE in 2h 34m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/54b195978f86465585befef4f86540a8 : SUCCESS in 1h 52m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bd64da35e79e4b4fa00a2911d4fe7a5d : SUCCESS in 2h 17m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f900f2466154475da7e2b15f240c1b71 : SUCCESS in 23m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f97eb51c7dad4f3a8ad5bba8ea30343e : SUCCESS in 1h 12m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0b67936655594630a1e44ea13d326de8 : FAILURE in 1h 18m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4beda34957534672857e86be26f0af83 : SUCCESS in 49m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8b0caaf67a594758947984f64cc6dd45 : SUCCESS in 1h 36m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d230235920334cf1a647199ff29c3176 : SUCCESS in 1h 39m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7566ce6d06674b04b5c405f005b1267d : SUCCESS in 1h 46m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fccd95737417403895eaf47763fd9451 : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":1},{"id":"795953aae377acbe0aeb730ef1a772450d13b378","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-30 08:56:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/537/.","accounts_in_message":[],"_revision_number":1},{"id":"4dd9092e8c263b1984a9d599405f286f8a1ecc58","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-05-30 09:01:14.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_05_2024_08_55_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"a5f11f81d1818059497fbe039a4c2fb766b2402c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-30 09:08:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp finger://managesf.services.openstack.netapp.com:7979/696518985f214095947adf548aa85c50 : RETRY_LIMIT in 2m 07s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/fdc593a13e9c43ec84b67998095d3bdb : RETRY_LIMIT in 2m 09s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/ab46dbafb87746f6b568139bdfbc6791 : RETRY_LIMIT in 1m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp finger://managesf.services.openstack.netapp.com:7979/b5dde39299934a35af18c069ca73fa58 : RETRY_LIMIT in 1m 36s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"bf8391df86fec71eb123c7a0d1a7e77162988bf9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-30 10:09:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_Unity_FC/f939e61/DellEMC_Unity_FC/2007 : FAILURE in 2h 05m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_Unity_iSCSI/5b03d89/DellEMC_Unity_iSCSI/2032 : SUCCESS in 2h 19m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7dada2db7673a8de0d4b8dbd7167b0711c870434","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-30 11:08:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_PowerStore_iSCSI/ad343bd/DellEMC_PowerStore_iSCSI/1398 : SUCCESS in 1h 15m 32s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_PowerStore_FC/277798f/DellEMC_PowerStore_FC/1581 : SUCCESS in 1h 01m 18s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_PowerStore_NFS/09598a4/DellEMC_PowerStore_NFS/1092 : SUCCESS in 1h 07m 44s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_PowerStore_NVMeTCP/3280f82/DellEMC_PowerStore_NVMeTCP/946 : SUCCESS in 2h 13m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"6e535092a32bb13fd4af0cdeff1ccdc5465bfe29","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-30 22:11:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_PowerFlex_3.6/bf55733/DellEMC_PowerFlex_3.6/1489 : SUCCESS in 1h 44m 19s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_PowerFlex_4.5/d9ae672/DellEMC_PowerFlex_4.5/403 : SUCCESS in 1h 49m 38s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/26/920326/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/6d48ecc/DellEMC_PowerFlex_4.x_NVMeTCP/342 : FAILURE in 2h 56m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"4ce0f265d3ab6f6b4c14e3bd09f5f28b6a6c5aff","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-31 21:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/26/920326/1/fc : SUCCESS in 0h 42m 56s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/26/920326/1/iscsi : SUCCESS in 0h 53m 52s","accounts_in_message":[],"_revision_number":1},{"id":"272427092d9dbfa552e5c9d3cdf4eb147adf7e46","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-06-01 15:33:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/920326-1/HPMSA-iSCSI/ : SUCCESS in 3:20:08\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/920326-1/HPMSA-FC/ : SUCCESS in 1:58:47","accounts_in_message":[],"_revision_number":1},{"id":"d2583d385c834af3a70e16e9e6342e51ce0e7aad","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-06-01 23:10:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/920326-1/Seagate-iSCSI/ : SUCCESS in 2:25:36\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/920326-1/Seagate-FC/ : SUCCESS in 2:19:00","accounts_in_message":[],"_revision_number":1},{"id":"af000aa97487fd5ee4a36da7f59ccd1bd48da4b4","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-06-02 08:16:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/920326-1/Lenovo-iSCSI/ : SUCCESS in 2:26:22\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/920326-1/Lenovo-FC/ : SUCCESS in 2:18:38","accounts_in_message":[],"_revision_number":1},{"id":"555295f6e0d47bb5df6aca8692a1f7556226d26d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-03 06:18:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac26b0464711a5151642a7880d1b232dae0087d0","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-03 07:20:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c6bd4cbf241b2af96a3078b08313105869c57dd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-05 10:51:50.000000000","message":"Abandoned\n\nthis test patch is no longer reqd","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b5dafb3f50307f5de27925731b75ee2ae1c08e9","revisions":{"3b5dafb3f50307f5de27925731b75ee2ae1c08e9":{"kind":"REWORK","_number":1,"created":"2024-05-30 06:14:07.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/26/920326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/920326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/920326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/920326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/920326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/920326/1"}}},"commit":{"parents":[{"commit":"b6236a56a917d8f801bf99205a3552a669bcf143","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6236a56a917d8f801bf99205a3552a669bcf143"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-30 06:27:49.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-05-30 06:28:26.000000000","tz":0},"subject":"DNM - For test only","message":"DNM - For test only\n\nDo not merge\n\nChange-Id: I64d16f0da2587f392ef676857950c1de897414d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b5dafb3f50307f5de27925731b75ee2ae1c08e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b5dafb3f50307f5de27925731b75ee2ae1c08e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
