)]}'
{"id":"openstack%2Fcinder~920991","triplet_id":"openstack%2Fcinder~master~If060af65900a8ed765e1bc70b365470fef82da0b","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-06-18 05:32:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If060af65900a8ed765e1bc70b365470fef82da0b","subject":"DNM - For test only","status":"ABANDONED","created":"2024-06-03 08:31:20.000000000","updated":"2024-06-18 05:32:57.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"457070c8ff10f8ce1808c49edfdcb5d10eeb935b","_number":920991,"virtual_id_number":920991,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-03 11:11:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-03 08:40:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-03 09:15:27.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-06-03 09:43:36.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-06-03 10:13:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-03 10:18:30.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-06-03 11:11:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-03 12:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-06-03 12:45:44.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-06-03 12:51:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-06-03 13:05:05.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-06-03 13:11:30.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-06-03 14:32:09.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-05 00:00:55.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-06-05 05:03:21.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-06-05 10:24:32.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-06-12 13:58:46.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"}],"messages":[{"id":"b30c5957105a64a68b3245d30c9e00a16febc4bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-03 08:31:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8f8938f60e1dafa446fa9df650dee1a6b1b8a97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 08:40:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/405a243fe79c490ea7c57f4d5ce09634\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15f9c4ef55594c648a0e5ef60b9fc8fd : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b054015c29b94127883e40738f1eb4db : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fbc43c32361a09938d3ca8f717547fef7898f5d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-03 08:52:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5755f3c6b6e8b1ebac2807fd83f0181285123249","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-03 09:15:27.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/719f6861883c4adbae04c8a8e5ccbd14\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3bd28b74c6974e1eb8ed4c9f18bdccc7 : POST_FAILURE in 40m 24s","accounts_in_message":[],"_revision_number":1},{"id":"aa5b550c0ea4f881a02a648f223db1062a1b511b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-03 09:43:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/920991 : FAILURE in 1h 09m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a26261bfd3f467e757393ee53066da59e34843a7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-03 10:13:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a156a943b2104f6d96a41e8589346641 : SUCCESS in 1h 31m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/00ea4c03bafc47e595ff8eeb322289cc : FAILURE in 1h 40m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d3633facf31965e429f8b850e262479dc4277492","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-03 10:18:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/550/.","accounts_in_message":[],"_revision_number":1},{"id":"ccd6ea6c4f594f21cecb51bfb6cb32d3b3587533","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-03 10:20:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/57/.","accounts_in_message":[],"_revision_number":1},{"id":"70872a433bd640a00f970e09a52e66fde3eca999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 11:11:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25f08323c7954668a4337a53b37e51bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0967202cb3455daf150bf1b55d8809 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdaa6eae7fab48c6af8ef01767f14494 : SUCCESS in 11m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3299619a89714519a7d1ed76d7dd620a : SUCCESS in 9m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a9be60ccaff4a698badc9688837ef65 : SUCCESS in 16m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/103b3259118c4f238eb931c6556bfcd6 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/3c09eca9ecb14047ad5b4d504499bf77 : SUCCESS in 48m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d88ea5fe2da40a1911ec6fb7ce7742a : SUCCESS in 1h 32m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57f072259b004d448508f5aaecfc3490 : SUCCESS in 1h 11m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb0c8ad683b449c09d9f47d610c6e416 : SUCCESS in 15m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e89ada8637c84091ba54c1d681aac1dc : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5d25513ced804c2d87c7914f8ae04b33 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/60dfea4699da4ac0b76f92690a57d21c : SUCCESS in 5m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fe7675d2d1c04554be5fd2df652279fb : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f9b634440ec43ea8f16c4271b4b480e : FAILURE in 43m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5bf0b1c33af94797937665175c5a5597 : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa145c43def2435c805eafa71ffec4ed : SUCCESS in 1h 54m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dbcad3c779d34ef398862b674a60f787 : FAILURE in 2h 37m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/29f98d5a3a644f0eb3e7e43300f48fd3 : SUCCESS in 1h 51m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2b508b29bf4244a29e087fd3e5662966 : SUCCESS in 1h 39m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/50449d5ef7684739b9b54b3c341ed39c : SUCCESS in 37m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e73c4a99cc734a38af2549a920cb8716 : SUCCESS in 56m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/56d6e605c5e44bbf87afec93550777ce : SUCCESS in 1h 35m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f04356f57b584313a84c80c8cfd928d1 : SUCCESS in 35m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df8eac7488814350824d49927eb6cb70 : SUCCESS in 1h 40m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/55d66ffee8ed4bcc95d7bb3dc2022e2c : SUCCESS in 1h 16m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4e1ab4a066f431682faac922b26f1fa : SUCCESS in 2h 00m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12bbb3be0b2141e9bd3960a51771ce93 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0cddcea4c5529c24ada2299cd86ea019bca92f37","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-06-03 12:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/91/920991/1/fc : SUCCESS in 0h 33m 17s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/91/920991/1/iscsi : SUCCESS in 0h 54m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9ba807a9c55e705483fa7c827aa1749f05ba534d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-03 12:32:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ef2b53264fc29dbe81a1f11089498eb23b98824","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-03 12:45:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/920991 : FAILURE in 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b65bc6a2fb804b8f48452179aea0ef6bd7972941","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-03 12:51:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/920991/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/a4de871/ : SUCCESS in 1h 27m 39s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/920991/1/upstream-check/cinder-tempest-plugin-ontap-nfs/58ec798/ : SUCCESS in 1h 29m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"529c46a84c9d5b7484eaf8d5f193f12a3f6bfe7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-06-03 13:05:05.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_06_2024_12_58_57\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"3483ec8a30b9aa60800971dccd5b9283b0d0f319","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-03 13:11:30.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/920991/1/check/DellEMC_PowerStore_iSCSI/9c8b69a/DellEMC_PowerStore_iSCSI/1406 : SUCCESS in 1h 34m 02s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/920991/1/check/DellEMC_PowerStore_FC/3d2bc65/DellEMC_PowerStore_FC/1589 : SUCCESS in 1h 16m 43s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/920991/1/check/DellEMC_PowerStore_NFS/60432ae/DellEMC_PowerStore_NFS/1100 : SUCCESS in 1h 10m 24s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/920991/1/check/DellEMC_PowerStore_NVMeTCP/84872b0/DellEMC_PowerStore_NVMeTCP/953 : SUCCESS in 1h 59m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"b778f40e7b2e9e2e3b0b43392ed4c090c37eed15","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-03 14:32:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/91/920991/1/check/DellEMC_PowerFlex_3.6/51ff233/DellEMC_PowerFlex_3.6/1506 : SUCCESS in 1h 49m 32s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/91/920991/1/check/DellEMC_PowerFlex_4.5/f7d831b/DellEMC_PowerFlex_4.5/416 : ABORTED in 6h 00m 01s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/91/920991/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/549c99f/DellEMC_PowerFlex_4.x_NVMeTCP/353 : FAILURE in 2h 55m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"6aeaf2993bb7dacd09fb3720aff6b2338fb153bc","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-06-05 00:00:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/920991-1/HPMSA-iSCSI/ : FAILURE in 3:28:52\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/920991-1/HPMSA-FC/ : SUCCESS in 1:57:42","accounts_in_message":[],"_revision_number":1},{"id":"d904676fcb6e533012734902a1e57044f44a7eef","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-06-05 05:03:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/920991-1/Seagate-iSCSI/ : SUCCESS in 2:28:05\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/920991-1/Seagate-FC/ : SUCCESS in 2:19:08","accounts_in_message":[],"_revision_number":1},{"id":"e8dec82fa596b4492233259f94b4dae8bde1f580","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-06-05 10:24:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/920991-1/Lenovo-iSCSI/ : SUCCESS in 2:25:15\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/920991-1/Lenovo-FC/ : SUCCESS in 2:18:07","accounts_in_message":[],"_revision_number":1},{"id":"7382a57aebd1a721b9e69c2000304b1fba5cee34","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-06-12 13:58:46.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/91/920991/1/20240612142108/NEC-V-ISCSI : FAILURE in 7h 45m 48s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/91/920991/1/20240612142108/NEC-V-FC : FAILURE in 8h 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"457070c8ff10f8ce1808c49edfdcb5d10eeb935b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-06-18 05:32:57.000000000","message":"Abandoned\n\nthis is patch not required now. closing it.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cc9b10fae97c9f59ba3c1675f0116bb6a81d3a4","revisions":{"8cc9b10fae97c9f59ba3c1675f0116bb6a81d3a4":{"kind":"REWORK","_number":1,"created":"2024-06-03 08:31:20.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/91/920991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/920991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/920991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/920991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/920991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/920991/1"}}},"commit":{"parents":[{"commit":"1ecfffafa6019cf6230a5af675e40c2a985dd6eb","subject":"Merge \"Update docstrings for methods related to driver initialization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1ecfffafa6019cf6230a5af675e40c2a985dd6eb"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-06-03 08:44:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-06-03 08:45:30.000000000","tz":0},"subject":"DNM - For test only","message":"DNM - For test only\n\nDo not merge\n\nChange-Id: If060af65900a8ed765e1bc70b365470fef82da0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cc9b10fae97c9f59ba3c1675f0116bb6a81d3a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cc9b10fae97c9f59ba3c1675f0116bb6a81d3a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
