)]}'
{"id":"openstack%2Fcinder~922888","triplet_id":"openstack%2Fcinder~master~Ib5e812a11e509ef0de9ad914239deecd987bcd74","project":"openstack/cinder","branch":"master","topic":"quobyte-unsupported","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-10 21:49:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib5e812a11e509ef0de9ad914239deecd987bcd74","subject":"Mark Quobyte Driver Unsupported","status":"MERGED","created":"2024-06-26 21:16:04.000000000","updated":"2024-07-13 04:01:32.000000000","submitted":"2024-07-10 21:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922888-quobyte-unsupported","meta_rev_id":"db98b8e3dec5479e90805ecfebac65b7ab3674a3","_number":922888,"virtual_id_number":922888,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-10 21:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2024-07-10 19:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-10 14:24:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2024-07-10 19:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 21:16:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-26 21:28:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-26 22:01:02.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-06-26 22:46:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-26 23:02:51.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-06-26 23:09:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-06-26 23:23:09.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-06-26 23:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-27 00:09:39.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-27 00:19:57.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-06-27 03:10:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-06-27 15:47:43.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-06-27 16:18:38.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-06-27 19:26:24.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-06-28 19:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-07-10 14:24:24.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-07-10 19:25:34.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-07-12 12:24:58.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-07-13 04:01:32.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"}],"messages":[{"id":"72c194e13f167d052aca17de965ceaddcee01590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-26 21:16:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab4ab95bf442bffc00d3ee4b4b223cbe3f773cba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-26 21:16:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"b3e691e618058ae13bc697f1545c19f7feffc72b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 21:28:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d2516014e954321b97c39f4f4021727\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e4e65788c4d41d3959569100fa29733 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8856db7941464f81a2630185b85d2e89 : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22a1e64761b31927b927ee73105c937d87679596","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-26 22:01:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/922888 : FAILURE in 43m 36s","accounts_in_message":[],"_revision_number":1},{"id":"09e555165c8cf4342db8edc0c0d1c884790be735","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-26 22:46:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/634a11335d044c5c9992d01d6ce3e167 : RETRY_LIMIT in 28m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6e37218d8c8149868e8774fd629520ca : RETRY_LIMIT in 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6b7ccd5a5672f5f3b03bbcc4841f1b1078014ee3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-26 23:02:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/643/.","accounts_in_message":[],"_revision_number":1},{"id":"d2c9cd5cb1622a6d138149e91da506aa235d9ff1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-26 23:09:55.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_Unity_FC/4555aab/DellEMC_Unity_FC/2141 : SUCCESS in 1h 27m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_Unity_iSCSI/d0a8e4c/DellEMC_Unity_iSCSI/2151 : SUCCESS in 1h 53m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"bee4b5f5a73a8d195624cf9438c288d092876f8b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-26 23:23:09.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3acd7f49b4bf4f17a81e20c241cbffba\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a37fc870385045cc82c7cdf06115874a : POST_FAILURE in 20m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6651bdfc5107aa54350fec5602079a5c07a43fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 23:38:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5845b587a6b4ea58428289079502f08\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f2ccc8ffee4f27b3aa7775a5734883 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/917ce3d3038645d9827d97b1029a5cfe : SUCCESS in 10m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/407072a506014731887ddcae1ea004a8 : SUCCESS in 10m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d243008eebcd4d2599e025afa5727058 : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a4ccfc03bc4ed5bb392d77738409b8 : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/0a0af5435d0c4458adc7bd3c484c50a4 : SUCCESS in 1h 05m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6ab5d5fb6d24def8ae6fd7d12253b84 : SUCCESS in 1h 33m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2a4f609d52c47ac843baea4ca41eb9e : FAILURE in 1h 01m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57deced8a65f4125affcd897b4452545 : SUCCESS in 17m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fd198c3da1b1454b8db6da651e01eca4 : SUCCESS in 15m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b075a71ca8514c57b6ae83b9f7780d02 : SUCCESS in 6m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/761a363f30ee4bed8a28be1305fe7e65 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17f72a58384f4e8caa572f4197025ddd : SUCCESS in 9m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/114756e2527b469c83f25bf937df2b4c : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c4f0f98cad7d4592bb7d607eb1be9395 : FAILURE in 36m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/836b6bb54ba6400bbcae9f58f4a8c13b : FAILURE in 4m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/79568ef8c324442b8d19e688f47a0bc2 : FAILURE in 1h 48m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0f1d4a83f147487ca895e3011410a312 : SUCCESS in 2h 18m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9512855b652e447c8b3000ffb2f4a76d : SUCCESS in 1h 44m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/046654a69a6040bfa591a7dc892b5280 : SUCCESS in 1h 36m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7639094425df4bcaa6e36d6442c918b1 : SUCCESS in 41m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e42fe642cdb452cb3c9da8f462aa6d9 : SUCCESS in 1h 09m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/515d9bf298cb4ea2ae25f45fde4948df : SUCCESS in 1h 26m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/21d8b5bb84c94478926aa52cdfafe946 : SUCCESS in 38m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ff9cad3456e447dab5fed3fb3a413be9 : SUCCESS in 1h 52m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/59b6e0ff76d448668ab8f905ea54dbfb : SUCCESS in 1h 36m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68e64da1b0c14eb8981c08ef78c9d524 : SUCCESS in 1h 49m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb7c37bdaee842dea2dffbc4abbf2b51 : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":1},{"id":"08d771fba69a50fbc0b8c698231e693d1b76cf2a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-27 00:09:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerFlex_3.6/60377cc/DellEMC_PowerFlex_3.6/1632 : FAILURE in 2h 53m 18s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerFlex_4.5/0ef12e1/DellEMC_PowerFlex_4.5/527 : FAILURE in 36m 30s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/1e46061/DellEMC_PowerFlex_4.x_NVMeTCP/479 : FAILURE in 12m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"3ed57d3a19d1c5bbe22a811723a887e1f4ecd2c8","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-27 00:19:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/922888 : TIMED_OUT in 3h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b81f5b8d2542545b8653540da286ff92b76c1721","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-27 03:10:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-922888-1/FC/ : SUCCESS in 2h 45m 55s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-922888-1/iSCSI/ : SUCCESS in 5h 54m 10s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"9ff27bba8b3f4335812017e13fb8badf1cf6e298","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-27 13:16:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86e10f141ecacf19f65eba522303072d9c42543f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 13:25:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/588d0cc48bbc49fe818c4fb18e695240\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28c195c7733c44b89653a0368ec7df3b : FAILURE in 3m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec17a41a79a2471f805eb95aaa97b8df : FAILURE in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd28871d6b35ecb1f60150cc253dad46e9f75671","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-27 13:39:13.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3befe0deff7042f5b3c0ed170497f67f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7213c0e50f9d4dfda71cd80683db9fb6 : POST_FAILURE in 19m 08s","accounts_in_message":[],"_revision_number":1},{"id":"340e932cd526ecd6883d1ef2ae67a02d978f85f7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-27 14:37:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/307d49dd7c8347f68d607a0d99a90211 : FAILURE in 1h 18m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d8105f788d2d48e4b8bb91ad18ae697e : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e0cc5cf5c9bcd9b5017e73d6e43d099269bd134d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 15:40:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85be0f0db17a4ea68ed52d40b18c0094\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37b05c7a74146d4a124dcecbfc0f0d0 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d73d1960b8be42699890ce1824454658 : SUCCESS in 11m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cba8c76fb1184fb7850c6b63d879e903 : SUCCESS in 11m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71416198a71e4f3d90f38b24a65a5480 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4751be0bbfcb44a3a2ac62f8ad1762eb : SUCCESS in 14m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/be20bb822f5a4a999eac812d77ba7be6 : SUCCESS in 1h 02m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/355479f961414759a441b1edd528258a : SUCCESS in 1h 37m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae12d77b108949c0862e812541b60527 : SUCCESS in 1h 01m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9d5538545084117bb3ae9a9a144213c : SUCCESS in 16m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8043a22a47204d92a215138056c47898 : SUCCESS in 16m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a2ec838b23ee467b9cdfa661002534ca : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1cf8266ee3564a178299989c33c2e4b7 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8bfc8da181f54f7daa4de55ce5f7784a : SUCCESS in 5m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a20020bf38da432889acd0c5edbf1658 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2a3a2b8b3664b338ac23b5b1acbf26d : FAILURE in 33m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa3d2de5551643a4a33aa74dc3732d83 : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cf85e43781a84f159c7a4a1ab334a6e8 : SUCCESS in 2h 10m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/afab2dfad7fe4026a713d481a0a591fd : SUCCESS in 2h 17m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/95acadf4a97a49d4b9ae8fcc2d3e127d : SUCCESS in 1h 42m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dca981b3c142474aa98daab9dd73f7fa : SUCCESS in 1h 50m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2f7e1fc7fd674d469e21da6ee73104b5 : SUCCESS in 36m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a07d2d56b6d64f90bfcfd3ac4e289b94 : SUCCESS in 1h 13m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/53fbb0339b0a455b9a339ba0d6fdf429 : FAILURE in 1h 46m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d834ee34b22d4a2aa5ed9f5d97ee5206 : SUCCESS in 40m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57b8e4591edc4066ae9c23012ee4d2ec : SUCCESS in 1h 29m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/742fad1e7382424b901af3b4898063e9 : SUCCESS in 1h 32m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f80ae7df6fa0458f92ae2cdd5f10bcfd : SUCCESS in 1h 30m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d0c964240854ef2b0cc654785ad3cea : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ab504921781166ad79845c94df505db1e08e730d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-06-27 15:47:43.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/27_06_2024_15_41_42\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"b95026e25242568e49a57eb8cbd75ae8bd935350","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-27 16:18:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/8475a1f325fe4737a12f8be3cb7370ae : DISK_FULL in 26m 49s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/c4af118a47384a77a9ad94042f113e3e : DISK_FULL in 36m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"086c9636c098aa3eb7fe18d95f6fb6acd4f24818","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-27 19:26:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerStore_iSCSI/0bea2b0/DellEMC_PowerStore_iSCSI/1476 : SUCCESS in 1h 23m 07s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerStore_FC/4bfccea/DellEMC_PowerStore_FC/1659 : SUCCESS in 1h 11m 08s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerStore_NFS/6d44234/DellEMC_PowerStore_NFS/1169 : FAILURE in 1m 36s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerStore_NVMeTCP/ab05509/DellEMC_PowerStore_NVMeTCP/1023 : FAILURE in 2h 17m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"d9663933661430ffaae8ba55de9534b2d907c95d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-27 19:45:07.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_Unity_FC/9b35060/DellEMC_Unity_FC/2148 : SUCCESS in 1h 23m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_Unity_iSCSI/c18e7ba/DellEMC_Unity_iSCSI/2157 : SUCCESS in 1h 40m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7aa65e53023a1035f35bcb4bd31ee05797f12c3f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-28 15:21:05.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerFlex_3.6/4c65d82/DellEMC_PowerFlex_3.6/1636 : SUCCESS in 2h 00m 34s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerFlex_4.5/91033bb/DellEMC_PowerFlex_4.5/532 : FAILURE in 2h 51m 33s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/88/922888/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/30fd502/DellEMC_PowerFlex_4.x_NVMeTCP/486 : FAILURE in 12m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"76d7d452f41d4dee4b4a59904d9b77a8583e5426","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-06-28 19:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/88/922888/1/fc : SUCCESS in 0h 43m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/88/922888/1/iscsi : SUCCESS in 0h 51m 54s","accounts_in_message":[],"_revision_number":1},{"id":"deac6651f222f39936e4bc8c7cfc62b29ac5ca36","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-07-10 14:24:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f75e76f6302ff3585967fb81de96053ee4006165","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-07-10 19:25:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"394d5ac106df2b70a1fc508b29e852622034a8c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 19:26:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7c7b972b1ab2a24f115f019bd415c85dbed1deb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 21:49:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf216d014144153b6cc2674e65e819c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a350ad34b9bb42f097ba6877fc94c902 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d6d2148bd284a8aa2b94db616f6dd68 : SUCCESS in 12m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2aab56cff054d42b3421b463a57bfef : SUCCESS in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0fdbf416014a479491ab60cba3d9c6 : SUCCESS in 11m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/eec697d2a731456b97ee2f3c2c0e6b4a : SUCCESS in 1h 09m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea639be54fed444f9203918ac1314619 : SUCCESS in 1h 32m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3da63df20a048f7b100177d3ab9540f : SUCCESS in 51m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1582fb7fc94345ec8680481f857d6cf8 : SUCCESS in 16m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8fd916545a2b416cb07903fa917eb244 : SUCCESS in 1h 11m 35s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/92a5abb6f5994c549450b828d0f1531a : SUCCESS in 1h 39m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44553ca9699643b4bee55821803edfca : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"af7691b646d89f010535ec3330f942f928fc084d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 21:49:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3ce3a678b5fff35d13f2e71f04133c50d40f3a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 21:50:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26b297335fc548db8b75c6cbd55d7c4f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b304f872f1b479d836e0b04dbce441c : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e7a2d3293714ec69ddc24a416bc6ec2 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9621824d1ed9a661c48bd2edecaac3368cb82ec3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-07-12 12:24:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/922888-1/HPMSA-iSCSI/ : SUCCESS in 3:13:51\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/922888-1/HPMSA-FC/ : SUCCESS in 2:29:53","accounts_in_message":[],"_revision_number":1},{"id":"db98b8e3dec5479e90805ecfebac65b7ab3674a3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-07-13 04:01:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/922888-1/Seagate-iSCSI/ : SUCCESS in 2:20:18\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/922888-1/Seagate-FC/ : SUCCESS in 2:14:14","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c1a77056672b35d51ef15242d59a2d81d8a7696","revisions":{"2c1a77056672b35d51ef15242d59a2d81d8a7696":{"kind":"REWORK","_number":1,"created":"2024-06-26 21:16:04.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/88/922888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/922888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/922888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/922888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/922888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/922888/1"}}},"commit":{"parents":[{"commit":"807066a484fca69dfcbb481b73e069e7dade0bbd","subject":"Merge \"Coerce booleans to integer values in paginate_query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/807066a484fca69dfcbb481b73e069e7dade0bbd"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-26 20:28:03.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-26 20:28:03.000000000","tz":-240},"subject":"Mark Quobyte Driver Unsupported","message":"Mark Quobyte Driver Unsupported\n\nThe Quobyte driver CI last reported on 2020-10-26 22:12:30.  As\nsuch, it can no longer be considered a supported driver.\n\nChange-Id: Ib5e812a11e509ef0de9ad914239deecd987bcd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c1a77056672b35d51ef15242d59a2d81d8a7696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c1a77056672b35d51ef15242d59a2d81d8a7696"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
