)]}'
{"id":"openstack%2Fcinder~922956","triplet_id":"openstack%2Fcinder~master~Idde5717d3251a0c8ead1c4f6a6a985f0aa39e923","project":"openstack/cinder","branch":"master","topic":"nfs-ci-probe","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-06-28 04:43:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idde5717d3251a0c8ead1c4f6a6a985f0aa39e923","subject":"DNM: looking at nfs CI job","status":"ABANDONED","created":"2024-06-27 14:53:46.000000000","updated":"2024-06-28 04:43:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c4062a893f248e9ddde73c4bde3b7317a4000c3","_number":922956,"virtual_id_number":922956,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-27 18:53:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-27 14:54:05.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-27 14:54:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-27 14:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-27 14:54:23.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-06-27 14:54:26.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-27 14:54:30.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-06-27 15:02:50.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-06-27 16:31:53.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-06-27 19:31:34.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-27 20:09:19.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-06-27 21:28:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"697d878b380ba96ccdc77a05eb2ffa1e9a28730a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-27 14:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"480bc4d49a9b317884ccc9173c454c97ff366270","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-27 14:54:05.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"fad714b0e0ecab792beac307f7ded71340bef1eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 14:54:09.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/cinder on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/cinder/.zuul.yaml@master\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"00e1303ed4804321b8db4159aac95a1534deddb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 14:54:21.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/cinder on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/cinder/.zuul.yaml@master\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"4eedebdc7556fc4e1507e84de01a65b0230df8d9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-27 14:54:23.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"da5c260b04f7101be5754ab7a27753d6a1795088","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-27 14:54:26.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"a6a1e568e949067fd3e482c51388fed33c8b2aa1","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-27 14:54:30.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"8b37528b7faab7cc4c5149ea1f9da58e78e00de1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-27 15:02:50.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"f551b50b0a92f5cc3e98f1160050843ae6c169d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-27 15:04:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2a67e1fe1d5b31690f12ed90b2879fb6183245d8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-27 15:05:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"92934669d63632a63cb164bd1e0f92cf647770f7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-27 15:27:21.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ff5d446295d44f3e97be705501be4fba\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e053e1dbf8f747e685783dde8e05830b : POST_FAILURE in 19m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6ad85af022af08cf571f98e4d7aed369020bd41b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-27 15:49:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/922956 : FAILURE in 40m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f91152ed303c776736274c45de890ec75e726548","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-27 15:50:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/922956 : SUCCESS in 44m 02s","accounts_in_message":[],"_revision_number":2},{"id":"37d991d978890bcd3259796a5a7f6b9e346f330b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-27 16:31:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/648/.","accounts_in_message":[],"_revision_number":1},{"id":"e9bbccdd6898644236e7dddf30117c116e9ff8b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:46:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/755af84bca6f4e8291cf74dd47df9014\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15bbae3c8d6e4111998a112c09150a0c : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa895dcf44f84168b4ce5b6f6d2bc821 : FAILURE in 10m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a5fc83046bf49bcb5439b2153854bfc : FAILURE in 11m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a43bfd05adf4e55833c9ccc70a5dbc1 : FAILURE in 17m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7bc8edd84906488492ac8f36726e4aed : FAILURE in 1h 14m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dd0abb225dd947cf8ccaf3a5ed03cf38 : FAILURE in 1h 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6b2dbafb88fa756cc7b8fe8909f6d2979400de0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-27 17:16:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0094d3a1b9a6219448e55626446470d94fc8606f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-06-27 17:17:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"c65526ab720fdee07e2a43943e3c1bb98e0ce4a1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-06-27 17:42:33.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/80a0861a206347d381a66454131f799a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e2a48cebaf3e43faaf282feb36202260 : POST_FAILURE in 20m 21s","accounts_in_message":[],"_revision_number":3},{"id":"62cbc535e4d17d45e01499701430bda470fd06b9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-06-27 17:57:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/922956 : FAILURE in 40m 06s","accounts_in_message":[],"_revision_number":3},{"id":"6b55d28d72c98fb081cb881941aa71a777e8850b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-06-27 18:01:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/922956 : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":3},{"id":"2277198601179e92467acecd9ff738bb47d58a97","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-06-27 18:36:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/027e642b68f947a389fd94606ce56d2e : SUCCESS in 1h 17m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0c573adfbb4c45b1967fb6cb62c7047e : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":3},{"id":"bf21889b27ce5ba33f027cf241f0f3badbc64afb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-06-27 18:49:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/650/.","accounts_in_message":[],"_revision_number":3},{"id":"0f6d289589eb619e7c8bdb5e2f3c9d40912c27db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 18:53:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/393d65f425e64501b44ab8159d1e60cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52a26f52541481d8dad5e70f5379db4 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c283a35f6d43485292fc2b9559b363ea : FAILURE in 10m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/322e969039874ab796bc59161608a282 : FAILURE in 9m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86125ebe9c874d399b907725cccd8e04 : FAILURE in 14m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0115bc8791244eeba0cf7111ae131383 : FAILURE in 1h 22m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/523a003695784028ba9ab24f1050947d : FAILURE in 1h 31m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"15a4e95c6f9c497a13ce4821e5ac25211b678a0b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-06-27 19:31:34.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/56/922956/3/check/fujitsu-eternus-dx-iscsi/3238ba1 : SUCCESS in 2h 13m 02s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/56/922956/3/check/fujitsu-eternus-dx-fc/09f0231 : SUCCESS in 2h 12m 22s","accounts_in_message":[],"_revision_number":3},{"id":"82e6165ec32cd20fa857177bffeeec0efcad06a7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-27 20:09:19.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-56-922956-3/FC/ : SUCCESS in 2h 21m 39s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-56-922956-3/iSCSI/ : SUCCESS in 2h 52m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"83105db30ee64e7daea5a44a6f04b68c64be5142","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-27 21:28:30.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/56/922956/3/check/DellEMC_Unity_FC/439a18a/DellEMC_Unity_FC/2149 : SUCCESS in 1h 34m 05s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/56/922956/3/check/DellEMC_Unity_iSCSI/43da15f/DellEMC_Unity_iSCSI/2158 : SUCCESS in 1h 43m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2c4062a893f248e9ddde73c4bde3b7317a4000c3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-28 04:43:08.000000000","message":"Abandoned\n\nNo longer needed.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3451c1a1a179c54b4c7387e618c53136c1646047","revisions":{"ccf392baeb50ff762bd245947e7420090755271c":{"kind":"REWORK","_number":1,"created":"2024-06-27 14:53:46.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/56/922956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/922956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/922956/1"}}},"commit":{"parents":[{"commit":"248436d1b5cf35e9c1b7ad1709c964e83856c2bc","subject":"Merge \"Add GMR to cinder wsgi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/248436d1b5cf35e9c1b7ad1709c964e83856c2bc"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-27 14:52:38.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-27 14:52:38.000000000","tz":-240},"subject":"DNM: looking at nfs CI job","message":"DNM: looking at nfs CI job\n\nChange-Id: Idde5717d3251a0c8ead1c4f6a6a985f0aa39e923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ccf392baeb50ff762bd245947e7420090755271c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ccf392baeb50ff762bd245947e7420090755271c"}]},"branch":"refs/heads/master"},"5b104eec8b65ed3e17a334984ef3b415453e1d43":{"kind":"REWORK","_number":2,"created":"2024-06-27 15:04:46.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/56/922956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/922956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/922956/2"}}},"commit":{"parents":[{"commit":"248436d1b5cf35e9c1b7ad1709c964e83856c2bc","subject":"Merge \"Add GMR to cinder wsgi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/248436d1b5cf35e9c1b7ad1709c964e83856c2bc"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-27 14:52:38.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-27 15:04:32.000000000","tz":-240},"subject":"DNM: looking at nfs CI job","message":"DNM: looking at nfs CI job\n\nChange-Id: Idde5717d3251a0c8ead1c4f6a6a985f0aa39e923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b104eec8b65ed3e17a334984ef3b415453e1d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b104eec8b65ed3e17a334984ef3b415453e1d43"}]},"branch":"refs/heads/master"},"3451c1a1a179c54b4c7387e618c53136c1646047":{"kind":"REWORK","_number":3,"created":"2024-06-27 17:16:41.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/56/922956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/922956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/922956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/922956/3"}}},"commit":{"parents":[{"commit":"248436d1b5cf35e9c1b7ad1709c964e83856c2bc","subject":"Merge \"Add GMR to cinder wsgi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/248436d1b5cf35e9c1b7ad1709c964e83856c2bc"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-27 14:52:38.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-27 17:15:55.000000000","tz":-240},"subject":"DNM: looking at nfs CI job","message":"DNM: looking at nfs CI job\n\nChange-Id: Idde5717d3251a0c8ead1c4f6a6a985f0aa39e923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3451c1a1a179c54b4c7387e618c53136c1646047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3451c1a1a179c54b4c7387e618c53136c1646047"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
