)]}'
{"id":"openstack%2Fcinder~923163","triplet_id":"openstack%2Fcinder~master~Id06d95128ca601c0a19e1a98060c28f11fa16abe","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-10 21:54:28.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-07-10 19:33:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-07-10 21:54:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id06d95128ca601c0a19e1a98060c28f11fa16abe","subject":"Deprecate GlusterFS backup driver","status":"MERGED","created":"2024-07-01 13:10:12.000000000","updated":"2024-07-14 11:50:40.000000000","submitted":"2024-07-10 21:54:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923163","meta_rev_id":"fd1f843957ff3fafb4681322c988dccadb8ad7c7","_number":923163,"virtual_id_number":923163,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-10 21:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2024-07-10 19:33:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-03 15:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2024-07-10 19:33:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-01 13:11:07.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-01 13:40:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-01 13:53:00.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-01 13:55:46.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-01 13:57:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-01 14:32:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-01 14:38:29.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-01 15:16:53.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-01 15:19:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-07-01 15:39:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-01 15:44:17.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-01 15:45:55.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-01 16:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-07-01 16:16:05.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-07-01 17:11:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-01 17:37:18.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-03 15:19:45.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2024-07-10 13:27:51.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-07-10 13:28:05.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-13 15:31:58.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-07-14 11:50:40.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"}],"messages":[{"id":"2fa5ba86cffd6a0274c7932f17d907a50325124c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-01 13:10:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce7472b8294a5f84f2833b72f80107bfc28a905e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-01 13:11:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"7228ba7f78cf65f40192c7ce7b18b7b6001e8d72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 13:40:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caaf55fbe68b477186c1914568b56bd1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4368e4c8105403ebcfaab65de9e87e3 : SUCCESS in 28m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f05cd65b377489981edf91d1aa242f5 : FAILURE in 3m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a61bae0d64ef394a1a1f3393db31878c4e62c8e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-01 13:53:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923163 : FAILURE in 41m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7d6e03ae36fd89d07c72ca0c68fb2b3ebfa40f77","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-01 13:55:46.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/80f4553be4a64ba38e4cd1730f63e40e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/338acb252e674409b314ac32d15aa5fc : POST_FAILURE in 21m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1abe52820ef233dd7b36ddeac1e06d9be9acada7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-01 13:57:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923163 : FAILURE in 43m 35s","accounts_in_message":[],"_revision_number":1},{"id":"76176f484ef23331ed968525e253bd39e8748794","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-01 14:32:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/654/.","accounts_in_message":[],"_revision_number":1},{"id":"64e4b16905da2286c368242de176c7aad332fc84","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-01 14:38:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/653473bc7fe44159afd34e984e4b1686 : SUCCESS in 1h 13m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/18fb348fb0da4ee98762f9f018cf676d : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b1cdd8906f98ee1ad0dd220f6a2957d59a0a901e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-01 15:16:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerFlex_3.6/3b673a0/DellEMC_PowerFlex_3.6/1639 : SUCCESS in 2h 06m 22s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerFlex_4.5/3daf7a7/DellEMC_PowerFlex_4.5/539 : SUCCESS in 1h 56m 28s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/8cc5c2a/DellEMC_PowerFlex_4.x_NVMeTCP/491 : FAILURE in 12m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d4e8a30ea8d08cdc7f2365ca3e8e7cf2faeaa9b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-01 15:19:31.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_Unity_FC/34b0d10/DellEMC_Unity_FC/2153 : SUCCESS in 1h 24m 49s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_Unity_iSCSI/adf5839/DellEMC_Unity_iSCSI/2162 : SUCCESS in 2h 08m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"98af0661b040bb2ee05960d43e56537b9bc2aca4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-01 15:24:37.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/169/.","accounts_in_message":[],"_revision_number":1},{"id":"c0f61c11142c7becfdefa0d7403494feacc61122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 15:39:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/688dc1bb06df49e7a0f3da2b70c08a9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e978e8ef8849889eac377614049c4c : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/890e6431cb4042c88b3c3283f7b9b521 : SUCCESS in 10m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7e4bd247d87452dabcb7b762cde5ddb : SUCCESS in 9m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aac1c237f04d41b7b5e68b3af21f731e : SUCCESS in 16m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62176051ff93461ea3fb1971947e6788 : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/c3f3b904050c41b1a05e6b7ac6c8574f : SUCCESS in 56m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/25d63bd45d5348d8a5154b2eeb9964f0 : SUCCESS in 1h 36m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59b51a162b1f46708046a0906830401b : SUCCESS in 1h 04m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c611306c8ef24895888820bcb94259aa : SUCCESS in 16m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0ffe3e4c4e4e46e494cb0157a44ae7f3 : SUCCESS in 14m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/410193490f85412f9ac71366ee6d5a3b : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c00cd5aebcd48cf854869820a0b4526 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/017cc15a53da4649bae98da221cfc967 : SUCCESS in 5m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4205747f63534f0099997112c6e0f9a6 : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8e908a4a9a5445388cede78741d459ff : FAILURE in 34m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd05cab67622417797c36e28e2c92e6d : FAILURE in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d157ce2d021487f98035dafb17f756c : SUCCESS in 1h 46m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/34d0ae1cd6694b4d93a99662ce6c96a0 : SUCCESS in 2h 17m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4023b95533cf4579b871c20faa26d229 : SUCCESS in 1h 28m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f70c6c8cd64c41cd9ddcdbcd75f0efc5 : SUCCESS in 1h 36m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dc5949bb5f8341cc936af30db9ffc753 : SUCCESS in 25m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c3c74256c0f848b8af89f6a7cdddde9f : SUCCESS in 56m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b8a39bbfeda849348c4359e1bf0fe708 : FAILURE in 1h 50m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/670577d0744d4f6f9cccc348ccd1b12e : SUCCESS in 42m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/009e7fea5f6f4113aa655376e9f9828a : SUCCESS in 1h 38m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d64f7f7811f1471b95a7695278aa214b : SUCCESS in 1h 08m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae1b9fe7f18345e5bf35723309f6c3bc : SUCCESS in 1h 29m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f1008aab6f34561b59ce14c74d74375 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"218926b918d77bb9d8ec71bf923b6368f926c437","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-01 15:44:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/63/923163/1/check/fujitsu-eternus-dx-iscsi/9244f6c : FAILURE in 2h 33m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/63/923163/1/check/fujitsu-eternus-dx-fc/f304768 : SUCCESS in 2h 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6c306348c8e780c2cd72cdabaf624d9d070db6d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-01 15:45:55.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_07_2024_15_40_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"f9add1f1b4924f6db2627317a72bebfcec51d99c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-01 16:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/63/923163/1/fc : SUCCESS in 0h 42m 16s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/63/923163/1/iscsi : SUCCESS in 0h 48m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d082cec6933fff03f25cbb52f8b9aab2e8293a30","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-01 16:16:05.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-923163-1/FC/ : SUCCESS in 2h 45m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-923163-1/iSCSI/ : SUCCESS in 3h 5m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"79ff4cc1a44dac97651b44bc052762be2580640e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-01 17:11:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/923163/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/0a8999c/ : SUCCESS in 1h 15m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/923163/1/upstream-check/cinder-tempest-plugin-ontap-nfs/8b9a6d9/ : SUCCESS in 51m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"8796d642101c6c5b50eb837468198c383f4b51f4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-01 17:28:19.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_Unity_FC/f698461/DellEMC_Unity_FC/2154 : SUCCESS in 1h 30m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_Unity_iSCSI/3280607/DellEMC_Unity_iSCSI/2163 : SUCCESS in 1h 48m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"407d264fa0268d0e43cd7ffff2c2c9af8689c489","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-01 17:37:18.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerStore_iSCSI/33edd95/DellEMC_PowerStore_iSCSI/1478 : SUCCESS in 1h 44m 41s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerStore_FC/20ece45/DellEMC_PowerStore_FC/1661 : SUCCESS in 1h 14m 36s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerStore_NFS/cb24ecf/DellEMC_PowerStore_NFS/1172 : FAILURE in 1m 36s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerStore_NVMeTCP/82aab5b/DellEMC_PowerStore_NVMeTCP/1026 : SUCCESS in 1h 57m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"a36d04ecfb845ff885931b633bf17b6a833ecd8e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-01 17:45:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerFlex_3.6/b45e9f7/DellEMC_PowerFlex_3.6/1640 : SUCCESS in 1h 54m 56s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerFlex_4.5/fc1e57a/DellEMC_PowerFlex_4.5/540 : SUCCESS in 2h 05m 31s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/63/923163/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/a61b563/DellEMC_PowerFlex_4.x_NVMeTCP/492 : FAILURE in 11m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"b8b0116b929c92461090d76b345ff20e5a89c596","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2024-07-03 15:19:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af1e8ab9eb7f583780d7a4c59f40bafb692ea4e7","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-07-10 19:33:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51c1120e1aa74ccf7304d42cf9177311bbaeacc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 19:34:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bbfd7a1f479bb543ba507ae63d44f50ef2a9021","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 21:54:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff58e5f587064474b2b886972d7062dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e26dbf4072f94c0093e7b680efb3339d : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/632384f54c66488aa3a38335039dcd3e : SUCCESS in 12m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5db82eacc1884d3e8c918733a14aa42d : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31da265a7370452e9d32b307a30a72a2 : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/5d4f8af66049405fb6ee2f21a730ce81 : SUCCESS in 54m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c19a90c6c298419da5086a44c0eeb1d2 : SUCCESS in 1h 09m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25e3444fd8d643f0bead887b85bb01d3 : SUCCESS in 58m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee93215ed63e4026a93a67ef54201245 : SUCCESS in 18m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2ff3064c3a9b4493b64a52377d83e3b3 : SUCCESS in 1h 13m 46s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/590bba0224d546e28cfe6dc445cf81a2 : SUCCESS in 1h 44m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53a0a0f886404a9c95d1a3393b063b22 : SUCCESS in 43m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0325b003d47a142811cb7d038369345e5f01df55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 21:54:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a24b757583c42b185daf6a1e0b056d2cd486feaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 21:55:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c5dfe2ef914d1f94acd82ac94d01de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77523a225374fceb43bba1cd427214a : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f59254fee17e4fc8bbeb67d857b5e2df : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"ee1ceca763da880aeb354e2431b9551396166ee3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-07-13 15:31:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/923163-1/HPMSA-iSCSI/ : SUCCESS in 3:16:24\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/923163-1/HPMSA-FC/ : SUCCESS in 2:29:43","accounts_in_message":[],"_revision_number":1},{"id":"fd1f843957ff3fafb4681322c988dccadb8ad7c7","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-07-14 11:50:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/923163-1/Seagate-iSCSI/ : SUCCESS in 2:19:18\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/923163-1/Seagate-FC/ : SUCCESS in 2:14:15","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3726afca561c9a81b1c2af1fb923a6e7484e7f64","revisions":{"3726afca561c9a81b1c2af1fb923a6e7484e7f64":{"kind":"REWORK","_number":1,"created":"2024-07-01 13:10:12.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/63/923163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/923163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/923163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/923163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/923163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/923163/1"}}},"commit":{"parents":[{"commit":"27ce7a167992cf8d86bce2b345929aecae41e4d6","subject":"Merge \"StorPool: create_cloned_volume() improvements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27ce7a167992cf8d86bce2b345929aecae41e4d6"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-07-01 13:01:42.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-07-01 13:06:21.000000000","tz":-240},"subject":"Deprecate GlusterFS backup driver","message":"Deprecate GlusterFS backup driver\n\nGlusterFS development has effectively ended, GlusterFS\nis reaching EOL in Red Hat Storage, and is not being\nincluded in CentOS Stream 10.\n\nMark it for deprecation with the plan to remove it in\n2025.1.\n\nReference:\nhttps://access.redhat.com/support/policy/updates/rhs\n\nChange-Id: Id06d95128ca601c0a19e1a98060c28f11fa16abe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3726afca561c9a81b1c2af1fb923a6e7484e7f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3726afca561c9a81b1c2af1fb923a6e7484e7f64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
