)]}'
{"id":"openstack%2Fcinder~923190","triplet_id":"openstack%2Fcinder~master~If569d72ab754c5ea5cd8ee2b5fe09b3d65bdef9e","project":"openstack/cinder","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"If569d72ab754c5ea5cd8ee2b5fe09b3d65bdef9e","subject":"Imported Translations from Zanata","status":"MERGED","created":"2024-07-02 03:26:01.000000000","updated":"2024-07-16 05:28:51.000000000","submitted":"2024-07-07 14:00:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923190-zanata/translations","meta_rev_id":"cf9e8d17c20043ea68352cb4b1c80cd86b18de27","_number":923190,"virtual_id_number":923190,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-07 14:00:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-07-07 13:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-07-07 13:37:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 03:28:42.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 03:37:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 03:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-02 04:02:21.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-02 04:09:24.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-02 04:10:46.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-02 04:16:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-02 05:01:11.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-02 05:40:58.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-02 05:47:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-07-02 05:52:00.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-02 06:22:31.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 11:13:13.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-07-07 10:19:42.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-07 13:37:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-08 03:16:11.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-07-14 17:07:22.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-07-16 05:28:51.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"}],"messages":[{"id":"ddef61641c96a46f9b03562bf482f55b8d590f1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-02 03:26:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87327190bf95ac92c9e5671d1d9ef27821d80f1b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-02 03:28:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"a2107ad232608de05a799d85cb38da468858929b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 03:37:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ec97bf06fd84db4a5bf9c6e8caf5f83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6fdb326dad0457ba1faf8359829fbdb : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1b6b1cde410041fcbe81454e664059b4 : SUCCESS in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b34b503aa9acdbaae0ea042e7cbad47b0d6ac206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 03:56:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cb7ba6f280749baabaaa22404f6ef27\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d47e8d8c07d48e59be395d918cabe97 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fae4d74e79474f02932dcfb31d2cc76c : SUCCESS in 9m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b94a5e7eb9546628e5b45d09085964c : SUCCESS in 11m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ccddd92dadb450fafc560eb05f839b1 : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db57e7e9d1ec4081884df1e4ad9b69b8 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/210c910aac16417880c401b852a6a0d4 : SUCCESS in 13m 20s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/75d7a8c32b7d4a1baf4c24b1bc32ab66 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"87bbc2aebc00c0cf79578eb472119452c1974f76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-02 04:02:21.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_07_2024_03_56_18\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"c22c600d8de4f7295c0921be98070edcb3f3febe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-02 04:09:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923190 : FAILURE in 42m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b9a14d8241990ab6398cdc9e20511cd1edbd433e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-02 04:10:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923190 : FAILURE in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f3b436bb162587f075cd4331411bd37f91825129","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-02 04:16:52.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ee17675386a045d9bdd4cabd2fb395e2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ae9745a9bf204a668521f7cf0b5741e3 : POST_FAILURE in 20m 12s","accounts_in_message":[],"_revision_number":1},{"id":"53bcacc4466684f9db0f250ba6af4adcf639d0d3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-02 05:01:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/923190/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/8e6f744/ : SUCCESS in 55m 10s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/923190/1/upstream-check/cinder-tempest-plugin-ontap-nfs/ecfecf8/ : SUCCESS in 55m 07s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"668d5d041c713b8c7fab9f81d75b004516527532","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-02 05:40:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/658/.","accounts_in_message":[],"_revision_number":1},{"id":"9cda2a2f198963baa770080a570f81cf83ec9ffd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-02 05:47:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_Unity_FC/891c0f8/DellEMC_Unity_FC/2159 : SUCCESS in 1h 33m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_Unity_iSCSI/35da05c/DellEMC_Unity_iSCSI/2168 : SUCCESS in 1h 35m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"fd965ca3114d9a5c4c97d6f0376657d6645baa20","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-02 05:52:00.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_PowerStore_iSCSI/ef2ae65/DellEMC_PowerStore_iSCSI/1480 : SUCCESS in 1h 27m 50s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_PowerStore_FC/c20ac10/DellEMC_PowerStore_FC/1663 : SUCCESS in 1h 13m 55s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_PowerStore_NFS/4ad6620/DellEMC_PowerStore_NFS/1174 : FAILURE in 1m 37s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_PowerStore_NVMeTCP/f463337/DellEMC_PowerStore_NVMeTCP/1028 : SUCCESS in 1h 55m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"6152aa3b0be1d2c8f15c236c7b9e39f526553007","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-02 06:22:31.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_PowerFlex_3.6/040a3fb/DellEMC_PowerFlex_3.6/1645 : SUCCESS in 1h 56m 15s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_PowerFlex_4.5/194d0af/DellEMC_PowerFlex_4.5/545 : FAILURE in 2h 02m 03s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/90/923190/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/fd25c2d/DellEMC_PowerFlex_4.x_NVMeTCP/497 : FAILURE in 10m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"e638ba334d6eb59b4de70540966dc22838b4d748","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-02 11:13:13.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-923190-1/FC/ : FAILURE in 7h 46m 44s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-923190-1/iSCSI/ : SUCCESS in 4h 4m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"f829d2bd55dc0d41e040e15955dd665c19338b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-04 03:41:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4292241308285a6a964ef6a46a1e73153fa0d818","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-04 03:44:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"6ff2a240fa1bba94f5c4fe209e73028155e1ffac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 03:54:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0687029dff04bfe96ddec178739857e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01e4da3069bb4512af16e566f6fbe033 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47b1bb19aa59467ab3f6ff6d23dfad11 : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b25f15ce37d939e364d0c24573e477fa8ea520c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 04:06:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faaa6ba2a5f3482dbaa7e8566840d6eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b8b15aced547488ec3fca041e0f3d0 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ecad83616054e88b8af3bdeb0bf2cab : SUCCESS in 11m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e52e7d291f64b3eb1cc856341253ebd : SUCCESS in 9m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6068e9bf341f4e8693957e4882171407 : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a276e8b6a2ff47d58a37c42c0fa36744 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42af0c68d4534dd4b8c3df524d5d2707 : SUCCESS in 18m 46s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47442a73a56144f5bc493d4f287a6a98 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"5efea2cdc725572282004e0378d5f3382b1aea34","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-04 04:06:43.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b1b8dfc7182244ad9b99c742f2e658d7\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/dd71490c4d194f08886fe62c4043f8c5 : POST_FAILURE in 21m 54s","accounts_in_message":[],"_revision_number":2},{"id":"85ade04247a2ab93bb376cb9568f6797140d4c1c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-04 04:22:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923190 : FAILURE in 39m 47s","accounts_in_message":[],"_revision_number":2},{"id":"35c53de5bc1a6e1f3299e27c3678009326e6dfa8","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-04 04:24:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923190 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b1e1aac3ce3e107b232dc93c797f0401409ee5a8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-04 06:12:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/923190/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/9f6c0ad/ : SUCCESS in 1h 36m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/923190/2/upstream-check/cinder-tempest-plugin-ontap-nfs/5675caf/ : SUCCESS in 1h 35m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7ff177107a572cf7a3581f635fe19d44de92757f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-04 06:31:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/680/.","accounts_in_message":[],"_revision_number":2},{"id":"cd4bde54d3dafeaf0450601ac1d83d3bf18af9e8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-04 07:19:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_PowerFlex_3.6/c771354/DellEMC_PowerFlex_3.6/1661 : SUCCESS in 1h 58m 44s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_PowerFlex_4.5/4971685/DellEMC_PowerFlex_4.5/561 : SUCCESS in 1h 40m 19s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/43e9af7/DellEMC_PowerFlex_4.x_NVMeTCP/513 : FAILURE in 11m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"4e0d8d7bd8890101899409349c3e228a9b5e5d46","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-04 07:29:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_Unity_FC/73f632f/DellEMC_Unity_FC/2175 : SUCCESS in 1h 32m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_Unity_iSCSI/4675a7b/DellEMC_Unity_iSCSI/2184 : SUCCESS in 1h 55m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"6504e820c069e2784bb1a0601460bbbda6807dc0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-04 07:57:50.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-923190-2/FC/ : SUCCESS in 3h 57m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-923190-2/iSCSI/ : FAILURE in 1h 17m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"a24bbe6dc71f84803755856c1a7dc728ed84b6ac","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-04 08:03:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_PowerStore_iSCSI/fc0e9a3/DellEMC_PowerStore_iSCSI/1486 : SUCCESS in 1h 24m 58s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_PowerStore_FC/ef2d898/DellEMC_PowerStore_FC/1672 : ABORTED in 2h 04m 10s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_PowerStore_NFS/4c76ef5/DellEMC_PowerStore_NFS/1179 : FAILURE in 1m 35s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/923190/2/check/DellEMC_PowerStore_NVMeTCP/b395616/DellEMC_PowerStore_NVMeTCP/1033 : SUCCESS in 1h 50m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"e3214e06493a543803f71a3cabd4fc288b2a3dd4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-04 10:56:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/180/.","accounts_in_message":[],"_revision_number":2},{"id":"c981ad93f8d85a8a6970c4e504a251b71b36691a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-07 03:06:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"75b824a13b45dfb2ee23de4badc6a405ddbf6164","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-07 03:07:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"3c229802995d625c2194b1305a368441ff098079","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-07 03:07:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"3d9b820349a2051f075426a5664365803422c35c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-07 03:07:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"e11f64334b357cf6eda63885fd3d5f3cf1aaafd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 03:11:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a19565bc402a4c399ec1a8a5b036bd04\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c374ce1d9f94a6a92bdcf5c6aa7ac3c : FAILURE in 4m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c21ed1e4be3340d092d46a3dc7ace235 : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"90fa814d8b5784c4ce8b8e40609f94bbabc2cac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 03:28:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3733cb61ba464e1782109a00c12dd53d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a01bd9e2e9459ea9e12e8102f2ffef : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba762c77d5a64a89ad42518fb7ef7b7c : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/102cb90988eb4228958ed3acf160941f : SUCCESS in 10m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17c5d8b64b534f7294caf1f26e8bc9f7 : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483c454f4e0a4adbb04eb8066bc781e7 : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c1dd1945e9b4493b149c9cb3a554685 : SUCCESS in 17m 29s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/63d92ef1dbb84acf824cedf208e3132e : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"3a611f2beac9b9acfe5a89ef1c705987999f060f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-07 04:48:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/686/.","accounts_in_message":[],"_revision_number":3},{"id":"55ff78b7346e89c525a443af5742e345587160c2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-07 04:48:53.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6da2ab4f450246a1a56f06524427505f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8f4bacf1f66241f8b9b52ee0abe44cb4 : POST_FAILURE in 19m 50s","accounts_in_message":[],"_revision_number":3},{"id":"8633146165539343289e3ad25ea4388f20976736","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-07 04:50:35.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/190/.","accounts_in_message":[],"_revision_number":3},{"id":"c7e5f04370951043489f727ee2e5e0527d0348ff","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-07 05:01:58.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_Unity_FC/59de045/DellEMC_Unity_FC/2181 : SUCCESS in 1h 27m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_Unity_iSCSI/c08fe47/DellEMC_Unity_iSCSI/2190 : SUCCESS in 1h 55m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2cc6ceccc5a5a641f95b8190a955749772fa5ebe","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-07 05:10:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_PowerFlex_3.6/0fd8309/DellEMC_PowerFlex_3.6/1667 : SUCCESS in 2h 03m 19s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_PowerFlex_4.5/024d995/DellEMC_PowerFlex_4.5/567 : FAILURE in 2h 02m 53s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/63548ad/DellEMC_PowerFlex_4.x_NVMeTCP/519 : FAILURE in 11m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"3e50b5ee491cfef553e847321cf3fc6155aaf982","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-07 05:25:22.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_PowerStore_iSCSI/7e9d33a/DellEMC_PowerStore_iSCSI/1490 : SUCCESS in 1h 23m 00s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_PowerStore_FC/eb1759e/DellEMC_PowerStore_FC/1681 : FAILURE in 7m 19s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_PowerStore_NFS/aa73a94/DellEMC_PowerStore_NFS/1183 : FAILURE in 1m 37s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/923190/3/check/DellEMC_PowerStore_NVMeTCP/882f8bd/DellEMC_PowerStore_NVMeTCP/1037 : SUCCESS in 1h 56m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"b9caa97d76e90ae24734cc15a541ac4b0ee45190","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-07 06:20:33.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-923190-3/FC/ : SUCCESS in 3h 13m 47s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-923190-3/iSCSI/ : SUCCESS in 2h 4m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"aab8fedb30fd63b37995bdfafa5ed61e749d8b57","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-07 06:46:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/923190/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/4a7138b/ : SUCCESS in 1h 23m 12s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/923190/3/upstream-check/cinder-tempest-plugin-ontap-nfs/a1e58b5/ : SUCCESS in 1h 25m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0a5e472f50581d6d0b30a40589d2748af12f5742","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-07 10:19:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/923190/3/20240707132422/NEC-ISCSI : SUCCESS in 5h 15m 14s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/923190/3/20240707132422/NEC-FC : SUCCESS in 5h 55m 16s","accounts_in_message":[],"_revision_number":3},{"id":"81b45f8aecb8fba83911ff15850cca28043d096d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-07 13:37:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0030a7dab50048b046e360699cb7e103bd7b5447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 13:37:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"608b6df09c8840619773cc50460135cb92682470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 14:00:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7160df0af12b47e98f2e405de18fffd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e88ca9f6a6f49e28a961cdf23d34ab7 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95e91ef4474a4600aa00ec45cf76d092 : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2796f19ca7b4c4698cc3874d5d02b59 : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac8b51f326347feba9b4088739815ff : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02d57559c4fe4e47a41b5dfe19747277 : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":3},{"id":"e00ea68d0ae5c58b3be07410951f5e49ebbe4f95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 14:00:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"671b85e23bb5595c435f0a1acc8c76fdef9437bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 14:01:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0353e253b81547ff99e85aa07178586d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27841a4b49154298a0233be7a76e238a : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8dde761461d4df6a054103c13bff05f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3},{"id":"92f00d99e6b7b4278374fd1c4cc97133898d0b34","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-08 03:16:11.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/923190/3/20240707200449/NEC-V-FC : SUCCESS in 3h 25m 43s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/923190/3/20240707200449/NEC-V-ISCSI : FAILURE in 16h 11m 19s","accounts_in_message":[],"_revision_number":3},{"id":"161b4e6a915f3de6db2e83a61067c5e6aee066bb","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-07-14 17:07:22.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"6128756e34ce401367f204038c3d06995cfc8405","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-07-16 05:28:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/923190-1/Seagate-iSCSI/ : SUCCESS in 2:19:07\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/923190-1/Seagate-FC/ : FAILURE in 18:34:56","accounts_in_message":[],"_revision_number":1}],"current_revision_number":3,"current_revision":"9b35fc5fa80d95d310de8716309aa378e0e3d7ec","revisions":{"68a9fe82a2d19e6e5abb4d527452e3151d2430b3":{"kind":"REWORK","_number":1,"created":"2024-07-02 03:26:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/923190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/923190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/923190/1"}}},"commit":{"parents":[{"commit":"0f649dcedd48c3f9bcdcf8bc86a4a773c69c0c2b","subject":"Merge \"pylint: Skip two E1130 errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f649dcedd48c3f9bcdcf8bc86a4a773c69c0c2b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-02 03:25:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-02 03:25:59.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: If569d72ab754c5ea5cd8ee2b5fe09b3d65bdef9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68a9fe82a2d19e6e5abb4d527452e3151d2430b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68a9fe82a2d19e6e5abb4d527452e3151d2430b3"}]},"branch":"refs/heads/master"},"398b954a4babcbe8ded7a52f071e56ab4bb95a4d":{"kind":"REWORK","_number":2,"created":"2024-07-04 03:41:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/923190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/923190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/923190/2"}}},"commit":{"parents":[{"commit":"346562008e50d4d3b6696ab5d88d2207215943be","subject":"Merge \"Remove the rest of ietadm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/346562008e50d4d3b6696ab5d88d2207215943be"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-04 03:41:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-04 03:41:48.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: If569d72ab754c5ea5cd8ee2b5fe09b3d65bdef9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/398b954a4babcbe8ded7a52f071e56ab4bb95a4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/398b954a4babcbe8ded7a52f071e56ab4bb95a4d"}]},"branch":"refs/heads/master"},"9b35fc5fa80d95d310de8716309aa378e0e3d7ec":{"kind":"REWORK","_number":3,"created":"2024-07-07 03:06:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/923190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/923190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/923190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/923190/3"}}},"commit":{"parents":[{"commit":"d5cd6f473622e2977e299318adaf92bb774243a3","subject":"Merge \"Remove \"from __future__ import annotations\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5cd6f473622e2977e299318adaf92bb774243a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-07 03:06:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-07 03:06:20.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: If569d72ab754c5ea5cd8ee2b5fe09b3d65bdef9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b35fc5fa80d95d310de8716309aa378e0e3d7ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b35fc5fa80d95d310de8716309aa378e0e3d7ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
