)]}'
{"id":"openstack%2Fcinder~923200","triplet_id":"openstack%2Fcinder~stable%2F2024.1~I30f105ee619386f52bc907f5115c08e0fafb9e42","project":"openstack/cinder","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-07-23 17:22:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I30f105ee619386f52bc907f5115c08e0fafb9e42","subject":"HPE 3par: getWsApiVersion now requires login","status":"MERGED","created":"2024-07-02 08:34:44.000000000","updated":"2024-07-24 04:37:17.000000000","submitted":"2024-07-23 17:22:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923200","cherry_pick_of_change":921592,"cherry_pick_of_patch_set":1,"meta_rev_id":"3d7ce15383979f90495a4b556e0aa43b340644dd","_number":923200,"virtual_id_number":923200,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 17:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-07-23 15:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-03 15:14:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2024-07-12 03:13:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-07-23 15:21:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 08:35:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 08:56:32.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-02 09:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 09:11:04.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-02 09:11:55.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-02 10:21:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-02 10:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 08:16:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-03 10:01:29.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-03 15:14:43.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2024-07-12 03:13:08.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"state":"REVIEWER"},{"updated":"2024-07-23 15:21:46.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"666da8ace68cfef0e8db7c4fdc92c71c6f487650","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-02 08:34:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1d82ecc1a5545a639c79a50931aa2dfaf4ef28a3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-02 08:35:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"73249db7ebae833799559ee01c48dff93f189d36","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-02 08:56:32.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ff5ad139541f4587a2685d1da97d036a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b4b631d4e8a541228ba671ac477596db : POST_FAILURE in 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b1faec9838c6ca2ea04834b8f4650efce8879311","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 09:06:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd85396e09904e97b9f94a04fd6aba0c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/514a932f3b9a4b21a2db3f3c45a7a0ac : SUCCESS in 26m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/453e375bd0ec4637a3158cbc870faeb4 : FAILURE in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed07b1caee6c69b3d319d2c35d98d5dd369fd046","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-02 09:11:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923200 : FAILURE in 34m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f5784e0791e7d2391d549a481c55964495cb1bf7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-02 09:11:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923200 : FAILURE in 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3548fe17039942b7924d5c743e25a32d68d00e7a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-02 10:21:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/660/.","accounts_in_message":[],"_revision_number":1},{"id":"3422328543d29f01c7d0e478b8fa226f85e60969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 10:59:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9136a6c91040435db39e54c77c7344a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03cc2b1bac0d4358a0f15d4000fdeb0d : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8726a06ec79041f79a83b1caf86a97e7 : SUCCESS in 9m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cca4e96a607d43e38c6acad06bc9f18b : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d97b8b167c4f56a86fc5bd7413ccc1 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/e5e4697ad05f4f6ab54db4cc66045110 : SUCCESS in 1h 01m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8c66d1b7bcf943cb813ba073446a719e : SUCCESS in 56m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8fcefe40fe4449c84653ddb94bf96f0 : SUCCESS in 1h 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a70bb0c4fb0488199405d6f9ab563ad : SUCCESS in 1h 00m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee2b641d91134e0d8f9a7a2ff8e16822 : SUCCESS in 19m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6ead2f3b8dc54f138fddddf3a6a6a795 : SUCCESS in 15m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a03b38a960234b3c9e654a6495bd35ac : SUCCESS in 6m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8564f39b84fb4530b394213f465249d0 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c16419df24c04a4984dc69b4bbfb2edc : SUCCESS in 5m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/189053a7cce84de8aa604942bf009a5c : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/18a7c42ab490411784fcc7b296c9ec19 : SUCCESS in 31m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a955beacdd6485aa65608c6275d14af : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d79587c7f1b342658c1abfcf8ec0596f : SUCCESS in 2h 11m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a9f257e977134fc78e8a8bbb5193881d : FAILURE in 2h 24m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a018a85ac3e14af8ad8e033d3a23ed12 : FAILURE in 1h 52m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d4b1fecf70a046f89c489d0fd220509d : SUCCESS in 1h 48m 55s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c6bfd7068ef424a9c007a74d2a4ae80 : SUCCESS in 1h 09m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9e0ff557b8cd4688b1dfaab29e094d2a : SUCCESS in 1h 39m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6c1e3697d26e447bb7192381439dfccf : SUCCESS in 46m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3e1df32c203f442898ff1ad62d96e696 : SUCCESS in 1h 14m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c131b7127a524bdbac1fae939027a98a : SUCCESS in 1h 15m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ede078450b1549d3b86e3bafdb48389f : SUCCESS in 1h 37m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9382b0238b104f749c6b9948ea656e76 : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fc0fdf3d748184a28a6ac61bbf823ec9f6da362d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-02 12:48:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7485f5a7ada96b8cbb93c7bad57877a0817d884","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 12:54:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b3fab6709e6423587c6c4f5031eca71\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a407ff9f35a4df18ab2b2d46a5ade5f : FAILURE in 2m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f8e9a6c272ba4713956070f1f08f0449 : FAILURE in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4710047f9d960ef3d452efbe3c9cdf4cd15985d8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-02 13:16:18.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/99df55862ae84054ab96f47ba10de0ab\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/44af808812224386ab1eff51769219cd : POST_FAILURE in 20m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5229006ed9efbd7381f051ab9193cc96bc777fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 15:08:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ff9230c279e44a1895fffadcaaee776\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2fcca3ff3884907ace03d2e0135cd54 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd419d3572ee44c7b4c82239f1858e65 : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e17f6b8542846869c356038c3a4121e : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c16566babcd4479889ffc5af51f0af4 : SUCCESS in 11m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/7f79fed59d2e4cf483de841c4a23bc66 : SUCCESS in 1h 04m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1e5bc9a7b1df41ccafaae9d8fa7cf280 : SUCCESS in 56m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e924abd8a554266ade02ae72fb8b973 : SUCCESS in 1h 34m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5aab9f15cf5b4c22a1e9f242f5b0a448 : SUCCESS in 1h 13m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69db361881294dd5bac9b802e97b8ef2 : SUCCESS in 18m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cbd74a1f9c2d4206bfd1fff485bb9c56 : SUCCESS in 14m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/09787a095631483dbd208b2725cfc1f3 : SUCCESS in 4m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/29114ca50bdd49e2b4d7106447eee4d9 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7bdae29ecc9440fc9d63737bcc24cb56 : SUCCESS in 5m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cca9516b1eb34200b6ba55e1540f6062 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b3644449efec43c4aee5cc42b6db9c35 : SUCCESS in 36m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3ebd542c931f470a82460bc240f05048 : SUCCESS in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ec50985cdfb840498ca760a69102c7a3 : SUCCESS in 2h 16m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd680297a5b64cb58d2732549e2487b6 : FAILURE in 2h 06m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65b22bad90b34620aff90554f5198364 : SUCCESS in 1h 46m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9eb1d4c1e77c4c4f844ba9ab81da4f9b : SUCCESS in 2h 01m 41s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/585411b217474c5ab2d90966a7900975 : SUCCESS in 1h 19m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cf5997184424448380c6dedae58767f6 : SUCCESS in 1h 38m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/23ccef66fde04c9b8dbd2f386c4c0cbb : SUCCESS in 41m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2792185181c9425d8e506d3b69ffd0d7 : SUCCESS in 1h 26m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e03f163e01b5477d809ce78539e62c75 : SUCCESS in 1h 36m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3cc0ce932a0f40b683b348e86a5b68dd : FAILURE in 1h 42m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f65e7a40b014ff4b4b24802a1c00ef6 : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e1ac01bce5cd4ade07c051ac05ee9cdaa0809629","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-03 06:01:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c853585ae9c2ba9e68c7eda67160a27d27ef7ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 06:10:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5877e808ca5d4fbd89e0189839e24136\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fd151008fea40ccb9ea26fa8a1a9bea : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c7f938ab97c84398b43214b1dff65f2f : SUCCESS in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee21f7feac436f64c66b94a2eba0b40534d224c7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-03 06:35:55.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/50c4d7ebcd1c4e82b8390397c9245e29\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0823b972081842dcb2c7c479ee0651d4 : POST_FAILURE in 19m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0cdac616465d8c71d1fac91831d14dbf1f51b97b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:10:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87c702cf14734cfaafd46d0699b3f1e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18cb0f3d546e41a39b6b65f799e227a2 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46144489c71c4d23a204377fd83c092a : SUCCESS in 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3805fdb2a0084d4191f3dadc5a6afa2b : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da512e41b464e6cbe2422764db2d639 : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/2203167b44fb46bcb063c948d12c86b2 : SUCCESS in 1h 03m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0d72c93f5de6414baf0cdf21a9f78a68 : SUCCESS in 58m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b7455b2cb174a1f83d6b8f3543a27b5 : SUCCESS in 1h 24m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/885a7228efd14918a53ad00acec7e467 : SUCCESS in 1h 08m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ffca53b298f4fc08bd6e10277ccc2ed : SUCCESS in 17m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/43d6bad0574543e7aa9e231f3f11cf51 : SUCCESS in 15m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6ae1a86b62c24ca2981ee7af40334963 : SUCCESS in 4m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/51f05e17e2f540e59381b679aaedb098 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/04c1242bfee749d88207b2807f9dc092 : SUCCESS in 6m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ebff8b22bb404d3a8e26b09a318d8c55 : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/12752aadba6b4f33b869cafcb15f45da : SUCCESS in 46m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ce70c0dc4b84f27bfaf68a5f6caa59d : SUCCESS in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/026666fb00b946aabae8cf0051f2b097 : SUCCESS in 2h 03m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6148aa88ace847b185dee1ae93f03d32 : FAILURE in 1h 53m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/27ae5cb64e1b475885de95aa9b22949a : SUCCESS in 1h 29m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/71edcd6b29a441069af3953cb6809bb4 : SUCCESS in 1h 42m 31s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fea305d25b0a4189b30393fba2c4b1b7 : SUCCESS in 1h 14m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/70d19e80f8cb4ff59fa600069e9e138f : SUCCESS in 1h 40m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a5d4bf33f754407a94cf57166be556c : SUCCESS in 47m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5d0d61cb3627423f8bce0720aa9a533a : SUCCESS in 1h 23m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ab02658794dd450cae4970d702131031 : SUCCESS in 1h 28m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b50b6becd75e420fb9630dfea17dbd43 : SUCCESS in 1h 54m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68a8dbe415344aba833fdbc580502068 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a9c00546be97c11da149d29c041810682c39772c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-03 08:16:42.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_07_2024_08_10_23\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"197d1cb0b655595c27f3d6ccdd7c54e1881b4dac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-03 10:01:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/923200/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/6c84fac/ : SUCCESS in 1h 34m 03s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/923200/1/upstream-check/cinder-tempest-plugin-ontap-nfs/5d78bb0/ : SUCCESS in 47m 44s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"676313f01a0bfc03ecc2c40168d71a4d0ceb4496","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-03 12:25:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19452124f8dff56c4ebc91129ae100aad69667c6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-03 12:26:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"c0191a5dfc34e7f900b12947e97b552c6e7df68a","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2024-07-03 15:14:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0ac98661153d3c846d742f310f85269ee0ca6a04","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-04 03:58:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"364f4b3b9940672e396c4096523503ad7e7e359b","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-07-12 03:13:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2e267cd170ce1359b9b4af4090a4f4c380091de","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-17 06:21:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d3e2dae80a3377d70bb721c8094e48cbbf93d1a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-23 15:21:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af2e19da34dbe3e4426818d78ea9e9aa251fb68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 15:22:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f08fe17aebc91767e6973321d8ab3d0aad31872d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:22:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/677b3d0aa92d45c4aa4133af47cb747a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164f3121060f4aee97164ed6717872a7 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0a62276fe524c5086af2be5114ea0c8 : SUCCESS in 11m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d37bfc197784d7a9124d5d8e0b4cc96 : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/826b621b92c74890b7c4ae5701bd4144 : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/ee390a06a3c64af88008121787ab3cb9 : SUCCESS in 57m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2d969774b0d341d6aa0c96ecbb870f53 : SUCCESS in 1h 05m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9939d1149a154bda8c60cdab2fa60123 : SUCCESS in 1h 46m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8ff52c86531432dbf33a6dc0ebde2de : SUCCESS in 59m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5199fe6115af465ebda0e31c02158783 : SUCCESS in 20m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d6dd633e4706493988fcfee9a5115e41 : SUCCESS in 1h 16m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d8df8d84c9a4fdf8e35cb997703dc13 : SUCCESS in 1h 54m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0f56ed8daf94ad4ad503c11e08f9003 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4efa6c2c44ab71d202e80ff493a91d881a0c4283","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:22:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8c5a67ae13ab72a9311f44c605d351d653b70208","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:23:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e8decc8e8ec467da3e7f2f30b0b07a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b233fa332b44a61aa73b3b8522be788 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b835beaf075b4b12a7342ca09831ec2a : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1},{"id":"3d7ce15383979f90495a4b556e0aa43b340644dd","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-24 04:37:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ab8848a29ec266490b75af505cf3b7ddfaac17c","revisions":{"7ab8848a29ec266490b75af505cf3b7ddfaac17c":{"kind":"REWORK","_number":1,"created":"2024-07-02 08:34:44.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/00/923200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/923200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/923200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/923200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/923200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/923200/1"}}},"commit":{"parents":[{"commit":"0ff4262fba803152e94e32e0dc8e4a2e56fcb0f5","subject":"Merge \"Fix broken backup_swift_service_auth\u003dTrue\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ff4262fba803152e94e32e0dc8e4a2e56fcb0f5"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-06-08 11:52:30.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-07-02 08:34:44.000000000","tz":0},"subject":"HPE 3par: getWsApiVersion now requires login","message":"HPE 3par: getWsApiVersion now requires login\n\nEarlier the call to getWsApiVersion() worked without login.\nNow with new wsapi of 2024, login is required.\n\nThis patch makes call to client_login() before getWsApiVersion().\n\nCloses-Bug: #2068795\nChange-Id: I30f105ee619386f52bc907f5115c08e0fafb9e42\n(cherry picked from commit 1b07bee38643d4d4cf52fcb01d2e45318187b27e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ab8848a29ec266490b75af505cf3b7ddfaac17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ab8848a29ec266490b75af505cf3b7ddfaac17c"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
